)]}'
{"id":"openstack%2Fcinder~948123","triplet_id":"openstack%2Fcinder~master~Ie8b05afb3f4dc6142ad8415a3cbdeb2584e8f900","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2025-06-03 17:43:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie8b05afb3f4dc6142ad8415a3cbdeb2584e8f900","subject":"tox: Match doc8 ignore-path with flake8 exclude parameters","status":"MERGED","created":"2025-04-24 19:11:24.000000000","updated":"2025-06-22 00:00:43.000000000","submitted":"2025-06-03 17:43:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948123","meta_rev_id":"558aa76ad939b457a1e2dce1622fd7d110e8958e","_number":948123,"virtual_id_number":948123,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-03 17:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-04-25 12:22:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-27 13:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-27 13:50:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-24 19:23:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-04-24 19:40:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-24 20:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-24 23:48:58.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-25 03:09:52.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-04-25 12:22:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-04-25 23:18:05.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-04-27 03:50:34.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-03 15:31:39.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-05-03 15:38:30.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-05-03 15:45:48.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-05-03 15:52:58.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-05-27 13:50:29.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-05-27 13:50:45.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2025-06-03 14:31:39.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"}],"messages":[{"id":"80d5f49ec8179a1b70680b93ff1323b0d4da0c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-04-24 19:11:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4456f9877c202493f0cf55fbf6012f403f0ac26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-04-24 19:14:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3d34861b03119bdd94741fde83d6a47fb87f686","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-24 19:23:59.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/779/.","accounts_in_message":[],"_revision_number":2},{"id":"f86e1c79205dae10edf62b21726804ea8836c2c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 19:40:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7e2c2a352fa48c191264613c6ede72b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01e44152769e4dd5bd247a69ac264b27 : SUCCESS in 16m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a4999fe1d6247a9b5c3ddb1ca18a105 : SUCCESS in 19m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/457bdd5ed6e541f098c5ddfb8ea51020 : SUCCESS in 23m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f88cff7b2f002b8cd578f62e469d8f7ee4628dd4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-24 20:06:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/660/.","accounts_in_message":[],"_revision_number":2},{"id":"4c8229fff4d2aac6cfc9e91cdebbbcd67aebd2ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 20:39:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a63b009a3142a3984601ff612f7303\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dd52e83d9d99461a94503ff174348a3f : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42b961b18bcb4d7c8b4589f7c0a36275 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae1967d375db41688ccb0675c7168a54 : SUCCESS in 10m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2cf59e3f4e534aafa9f2f03dd96bffc2 : SUCCESS in 13m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06be2a231f7b43bc83e334e14ada90bf : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69e106cba53e4e5a8b218493533d3423 : SUCCESS in 14m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8ac79ac1a39449fc993e51d60d37dcda : SUCCESS in 59m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/780bd51a702f4e5aa1d8aa40bc3ee4d0 : SUCCESS in 10m 32s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c02506a3b31a4412bb471edac9f1220d : SUCCESS in 3m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f48930f379634f408730afd9208bff2d : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4b0de88fb6a04be7bb6f3f6af77a91cc : SUCCESS in 4m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/83b14983f8e34488a86b26655c09cf5b : FAILURE in 26m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f126fe3626434f6a8657a4d150e4a770 : SUCCESS in 7m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f8a72c537a84884bf540e80eb1131e4 : SUCCESS in 1h 22m 44s","accounts_in_message":[],"_revision_number":2},{"id":"279f98d036e6414e9d8afdd6ab80f188ae37cbea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-04-24 22:23:47.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bd74dd297cb3bec048e4dcf078e6ba19efb24b23","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-24 22:34:01.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/780/.","accounts_in_message":[],"_revision_number":3},{"id":"5b66116310fe0d6f8d335a5e05552ef87a6fbf57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 22:48:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ede667e85f384ee1b67ea788e5589dc8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5495332944e34fdf89ce36b3fe75ac97 : SUCCESS in 17m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99d778b0842441de91a21997a283534c : SUCCESS in 19m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/54ab496ea8134bfe92c6be8300404823 : SUCCESS in 22m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"14710362427c5119d92ec4a85af93a097f189f54","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-24 23:21:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/661/.","accounts_in_message":[],"_revision_number":3},{"id":"4ed0947078b7df7b999790217caa38bf0a696b02","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-24 23:48:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3bd32a7c8f7d48faa0ad5bb6e4f1dc6d : SUCCESS in 1h 13m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/52485a71045649d4a228bbc02204e184 : SUCCESS in 1h 23m 27s","accounts_in_message":[],"_revision_number":3},{"id":"0886c3c951ece2acef3158317a2aae1cfad7a58f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 23:57:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36108bc257fe4d9b92311987a77e83da\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e9599fb5ec3e41b2bb3626953cdc9860 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/096c2a632f9844e4af91bd32905e2da9 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9de976c7a4b146bf859391ec4d9bfaae : SUCCESS in 8m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e966f3c02cbd4b1a85b0e4e31c735748 : SUCCESS in 7m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ff1a878206e4db69eba4d6789f11e23 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccfa7a1077b846c59bc1143aa2b54baf : SUCCESS in 7m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d0c1198ff77b45d2b73e3a2281c3e666 : SUCCESS in 40m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7579d326bb248579b9b25b24c9af4f9 : SUCCESS in 10m 15s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f35c7502811b410cbd7872b50b4d7763 : SUCCESS in 4m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a3e02782005347909491f76c4787faae : SUCCESS in 5m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0790805d7f174c0eafccb724a2fbbc97 : SUCCESS in 4m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cb4026e85137434e9390da4afba008f8 : FAILURE in 38m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/52d89e3a5d814ac4b6af0ad8c5224d82 : SUCCESS in 3m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0d4577f4cf0d4ac48a8b55e63846ba7f : SUCCESS in 1h 17m 24s","accounts_in_message":[],"_revision_number":3},{"id":"8af39d6ff41be9d40fabc07c72553f9dbd14de72","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-25 03:09:52.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-948123-1/FC/ : FAILURE in 7h 58m 1s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-948123-1/iSCSI/ : UNSTABLE in 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"3e2ab6e149b8e079b0b0012045a9ff56b9e2b3ef","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-25 03:26:05.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-948123-2/FC/ : FAILURE in 8h 10m 56s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-948123-2/iSCSI/ : SUCCESS in 2h 14m 40s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"1b9bb646f09e1ebfd58dccd7371da922b4e45209","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-04-25 05:58:07.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-948123-3/FC/ : FAILURE in 7h 33m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-948123-3/iSCSI/ : SUCCESS in 2h 26m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"ba44cb1dfbb0b0289445f0cc5757c1760af2fd3a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-25 06:04:28.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/781/.","accounts_in_message":[],"_revision_number":3},{"id":"6da29d0281cbf12a4755c96d1182c90873250f64","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-04-25 12:22:37.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a61127e3ce1d1d6761d64e56f0ea51c3c20068c6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-25 23:18:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/23/948123/3/20250426040456/NEC-V-ISCSI : SUCCESS in 4h 05m 33s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/23/948123/3/20250426040456/NEC-V-FC : SUCCESS in 4h 13m 06s","accounts_in_message":[],"_revision_number":3},{"id":"2d6ab3e8596c473dcbd1c22075540bfafe5234ea","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-27 03:50:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/23/948123/3/20250427065608/NEC-ISCSI : SUCCESS in 5h 31m 40s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/23/948123/3/20250427065608/NEC-FC : SUCCESS in 5h 54m 23s","accounts_in_message":[],"_revision_number":3},{"id":"890d24020321476984c8e46d50e1a92185c4d55e","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-05-03 15:31:39.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"32c496f54ff5806a2cf0b187c8ed2417871342c9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-05-03 15:38:30.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"a2300c6dfbce502cd1823ed8774532827848e5be","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-05-03 15:45:48.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"b72009972111dbb1ed0c9cdbb8d2906e0f1a2f87","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-05-03 15:52:58.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"f74426a94c5439af6a8ab0109e029e08e5debe68","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-05-03 16:00:11.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"5557bbf1acbeb93deb4bb5d27b0acac2334b545f","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-05-03 16:07:27.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"9f43a6a8f50199de0fcca8b81186c6a79e760b4c","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-05-03 16:14:42.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"d719b68d809b637c1b920221da15292891a9137e","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-05-03 16:22:00.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"06832c09bf9908e6d2fba4b7096d813ead67643a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-27 13:50:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3bb7494bd73f84cccd68a22d8becfdf968b2e4bd","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-05-27 13:50:45.000000000","message":"Patch Set 3:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aed33e2d1ceddbd6aaa5a927e9a26f9379fcfb64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:50:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3cd121e7f9b3f11e060f3c3cf14223717da24621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 16:59:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0281701b1aa74d2a93d215c6fd55f95c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/82baabbef5674a068d4e463f93608367 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeb75568b4104961a099ea000d693ebc : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11212b0205284cb5b22db10d4c3b2c10 : FAILURE in 11m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e46ae44e3294fd5800d3227237e1c68 : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3766bc92937f459b88b33f913a3c117c : SUCCESS in 16m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/71f64cd11e6c4f2784aba772f4f0553e : TIMED_OUT in 3h 02m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cae0a9c904234260bb03efba284c75e8 : SUCCESS in 21m 24s","accounts_in_message":[],"_revision_number":3},{"id":"4c8bb9ce60638f36b82fb1ea79727ea4a705c240","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-05-27 22:31:28.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/4e286a4b2b3e4972afc8e3992da1b714\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/2bb466069adf4087ae4cae771e258c87 : FAILURE in 8h 40m 41s","accounts_in_message":[],"_revision_number":3},{"id":"2c9d2760e25e03941b92c82ccac6595836b73b2c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-28 22:07:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"35c3003d588cff6e0a53f0cbea8cfd261b6a4703","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 23:11:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a794e5d16d5c48658424098c28e73099\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da24db221981441482960354e3552a19 : SUCCESS in 23m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/734a9c40fa9d4990b0cc2300661dcff1 : SUCCESS in 24m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/92fc3bf08eb444d3a9d48cba89924457 : SUCCESS in 30m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b8910d6a75bc9f07aa1bd2da859a7f1b46f339d9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-29 01:10:47.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0c6677aaae4048139de2a030e169930c : SUCCESS in 1h 18m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cd76af2000724e248818c198292eab5b : RETRY_LIMIT in 17m 21s","accounts_in_message":[],"_revision_number":3},{"id":"bc3edd0b08608c4c6a93cd1258d311346d47f846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 01:11:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2658b46235e74f7abc4027d62b9f0114\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/40086175c7b14e59be81c7a5d7e57aa1 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2511660f80545b6a8d304e8cab68297 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d7c9ad6258542cda161bd72f89f4d65 : SUCCESS in 11m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/39220c5adc7542c5bf7030fc2e12338d : SUCCESS in 11m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e996f03c08904d458a47c9ca7e697588 : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a49746f0af5848d3b1f4c36d91faf6cd : SUCCESS in 14m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3b61765758c24aafb2a9e8a2ce972336 : TIMED_OUT in 3h 02m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfb109e4bf7946eb95b6e3907f3b14d4 : SUCCESS in 17m 26s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8cfbb16b415b4015a4267d28775f9626 : SUCCESS in 6m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/14dd078500884f4587c86c0865465406 : SUCCESS in 5m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5c44e1cbbc384788b53700b9fc68716d : SUCCESS in 9m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9ef637b19c946b0b75687dbacf7a18d : FAILURE in 37m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37f08f24e7044fca8c7ce76b21fcf8ba : SUCCESS in 5m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ecf58f2e9a58419d88830e01b960a741 : SUCCESS in 1h 52m 54s","accounts_in_message":[],"_revision_number":3},{"id":"be35c6f1d5104d6a581a11519aeb96af805eccb2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-06-03 14:31:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aa611d273325e4fce8711970b0bbaec5cc4f9a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 16:24:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69c9b6a3e25d4644bb052d9c6a0b26b5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2412b1b7a5a24b51b637336b8988fa70 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2b882ff1bce41fc8648af6c8bbbdfdf : SUCCESS in 7m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/121ff3e6e163429996cd4546936b8611 : SUCCESS in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eaa771f34c734cb5814714732c90a42c : SUCCESS in 9m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f7cb7479d8447bdba09a4ab0e79dfdf : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b664dd73c11d47d0b724e6815a0d912b : SUCCESS in 15m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b271397cce442d3992b46b6d077358d : SUCCESS in 57m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4905c7c887a4b65bb0c457f1fa4351d : SUCCESS in 24m 42s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6a2893573ae741a899e6933c56e32f2a : SUCCESS in 5m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a1aa9bb34680442b988e972fe74746db : SUCCESS in 6m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4ae1340e5230452fab2b57e87607244e : SUCCESS in 9m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7a9ce45f70644b5e945c98a9075c6675 : FAILURE in 34m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1f5dfb67345d4fa29f1e9b2ecd8c28ef : SUCCESS in 6m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/915f8b1fdc134d35bc1ec0900c9091d3 : SUCCESS in 1h 41m 40s","accounts_in_message":[],"_revision_number":3},{"id":"60e928f22821afaec252f72c8025818f8147702c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 16:24:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6c6383e5f18c3e8fc42de7abceb50baf93d1e5c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 17:43:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87d9ad7ca79a4c178b2369cb6f7ed52a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/32a51c870b6a4cbab35b798a9ccc3920 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a528d6a88cdd404f95b4c70aa6c07133 : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e3d1dd88ffd4d27a698909027149ccc : SUCCESS in 9m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b638a6b7dc4e43a1b1190ac3fe3bb694 : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03480e4035b64635a818386bfecde310 : SUCCESS in 13m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78a69c1e0b77456189d6fed59a44c5f2 : SUCCESS in 51m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5e13ca77f4742e8934e117b0365e227 : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":3},{"id":"8f3807ba9074262fb133320687b499f1419eed25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 17:43:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"37e07d46d48910f3b11fd32f493e2e4e50353a68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 17:44:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51496271107f42d398df8a925706234d\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/641d4e06c8e34cd8a44fbd4c546d6b20 : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de850ebc2434b9ea67695bd645d193f : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1b90a0118234661a4bd06ee7bd763fe : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3},{"id":"4196254f32075a7038a9be4e3d88716ef93906d1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-03 19:53:59.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8d024de838674d5e9a7374913035c4d8 : SUCCESS in 1h 16m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/14930b96166e4a4ca779d82447c009c3 : SUCCESS in 1h 23m 43s","accounts_in_message":[],"_revision_number":3},{"id":"fe905a9d03cd2fc98571498d22b68714f367939c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 22:32:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca3024c51a3641418cb19ad307928e01\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c2df1b2ede0b4f3cb1e1d75c05ade4db : FAILURE in 1h 01m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/20779728d639469f96fdd3174d866f1a : FAILURE in 52m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/10b61ef6de5c4e31b5bb06d34a835a7b : FAILURE in 53m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"22de0ef13b886f589c5c6fe7fceaff4a0f77172d","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-06-20 21:30:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/948123-3/Seagate-iSCSI/ : SUCCESS in 2:25:49\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/948123-3/Seagate-FC/ : FAILURE in 6:06:39","accounts_in_message":[],"_revision_number":3},{"id":"c75e64151b3f2f3b26d273730bf009192d3c93a4","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-06-21 06:23:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/948123-3/PVME-iSCSI/ : SUCCESS in 2:25:17\n- PVME-FC https://os-logs.seagate.com/dell-ci/948123-3/PVME-FC/ : FAILURE in 6:05:46","accounts_in_message":[],"_revision_number":3},{"id":"bb74e9005474df33ad9fe578aef531decdf1b0d4","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-06-21 15:16:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/948123-3/HPMSA-iSCSI/ : SUCCESS in 2:23:38\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/948123-3/HPMSA-FC/ : FAILURE in 6:06:28","accounts_in_message":[],"_revision_number":3},{"id":"558aa76ad939b457a1e2dce1622fd7d110e8958e","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-06-22 00:00:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/948123-3/Lenovo-iSCSI/ : SUCCESS in 2:26:44\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/948123-3/Lenovo-FC/ : FAILURE in 5:56:24","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8dbe8b8cda6ebb28175292f5a24f358252eec8db","revisions":{"f553028b8a8ddaad1509e20e538c4a3d0262b47d":{"kind":"REWORK","_number":1,"created":"2025-04-24 19:11:24.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/23/948123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/948123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/948123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/948123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/948123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/948123/1"}}},"commit":{"parents":[{"commit":"a0b8f270e60606cf5ae1fe97fd1abe6030bb10c9","subject":"Merge \"NFS driver: Fix fail creating volume with multiple snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0b8f270e60606cf5ae1fe97fd1abe6030bb10c9"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-04-24 18:48:25.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-04-24 19:05:15.000000000","tz":-180},"subject":"doc8: Match ignore-path with flake8 excludes","message":"doc8: Match ignore-path with flake8 excludes\n\nDifferently from flake8, the current tox doc8 setup isn’t configured\nto the exclude the commonly used directories ``.venv`` and ``.git``,\nwhich can be inconvenient in development environments.\n\nThis path includes ``.venv`` and ``.git`` directories to the\nignore-path parameter in tox, as found in flake8\u0027s\n``exclude`` setting.\n\nChange-Id: Ie8b05afb3f4dc6142ad8415a3cbdeb2584e8f900\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f553028b8a8ddaad1509e20e538c4a3d0262b47d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f553028b8a8ddaad1509e20e538c4a3d0262b47d"}]},"branch":"refs/heads/master"},"511ed9db9cd12e1270e7281db7c833dfbcde727e":{"kind":"REWORK","_number":2,"created":"2025-04-24 19:14:40.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/23/948123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/948123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/948123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/948123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/948123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/948123/2"}}},"commit":{"parents":[{"commit":"a0b8f270e60606cf5ae1fe97fd1abe6030bb10c9","subject":"Merge \"NFS driver: Fix fail creating volume with multiple snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0b8f270e60606cf5ae1fe97fd1abe6030bb10c9"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-04-24 18:48:25.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-04-24 19:14:32.000000000","tz":-180},"subject":"doc8: Match ignore-path with flake8 excludes","message":"doc8: Match ignore-path with flake8 excludes\n\nDifferently from flake8, the current tox doc8 setup isn’t configured\nto the exclude the commonly used directories ``.venv`` and ``.git``,\nwhich can be inconvenient in development environments.\n\nThis path includes ``.venv`` and ``.git`` directories to the\nignore-path parameter in tox, as found in flake8\u0027s\n``exclude`` setting.\n\nChange-Id: Ie8b05afb3f4dc6142ad8415a3cbdeb2584e8f900\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/511ed9db9cd12e1270e7281db7c833dfbcde727e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/511ed9db9cd12e1270e7281db7c833dfbcde727e"}]},"branch":"refs/heads/master"},"8dbe8b8cda6ebb28175292f5a24f358252eec8db":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-04-24 22:23:47.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/23/948123/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/948123/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/948123/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/948123/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/948123/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/948123/3"}}},"commit":{"parents":[{"commit":"a0b8f270e60606cf5ae1fe97fd1abe6030bb10c9","subject":"Merge \"NFS driver: Fix fail creating volume with multiple snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0b8f270e60606cf5ae1fe97fd1abe6030bb10c9"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-04-24 18:48:25.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-04-24 22:23:47.000000000","tz":0},"subject":"tox: Match doc8 ignore-path with flake8 exclude parameters","message":"tox: Match doc8 ignore-path with flake8 exclude parameters\n\nDifferently from flake8, the current tox doc8 setup isn’t configured\nto the exclude the commonly used directories ``.venv`` and ``.git``,\nwhich can be inconvenient in development environments.\n\nThis path includes ``.venv`` and ``.git`` directories to the\nignore-path parameter in tox, as found in flake8\u0027s\n``exclude`` setting.\n\nChange-Id: Ie8b05afb3f4dc6142ad8415a3cbdeb2584e8f900\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8dbe8b8cda6ebb28175292f5a24f358252eec8db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8dbe8b8cda6ebb28175292f5a24f358252eec8db"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
