)]}'
{"id":"openstack%2Fcinder~949483","triplet_id":"openstack%2Fcinder~master~I599167d0b63ff1d4cd0330b390a2dc33a4f06efb","project":"openstack/cinder","branch":"master","topic":"pbha_merge1","attention_set":{"37888":{"account":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"last_update":"2026-06-19 19:54:46.000000000","reason":"Someone else replied on the change"},"36800":{"account":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"last_update":"2026-06-19 19:54:46.000000000","reason":"Someone else replied on the change"},"37328":{"account":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"last_update":"2026-01-29 11:52:41.000000000","reason":"Someone else replied on the change"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-07-08 08:33:46.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-04-30 18:54:43.000000000","reason":"Someone else replied on the change"},"32074":{"account":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"last_update":"2025-05-12 10:28:29.000000000","reason":"\u003cGERRIT_ACCOUNT_37888\u003e replied on the change","reason_account":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"}},"37470":{"account":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"last_update":"2026-04-28 20:58:59.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"38067":{"account":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"last_update":"2026-03-31 06:20:05.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-09-10 10:27:46.000000000","reason":"removed on reply"}},"hashtags":["ibm-svf-cinder"],"change_id":"I599167d0b63ff1d4cd0330b390a2dc33a4f06efb","subject":"[IBM SVf driver] PBHA Volume Extend Support","status":"NEW","created":"2025-05-12 10:24:34.000000000","updated":"2026-07-13 07:43:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":32,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"8e7bd7eb20c030107ebe297ef5344509810c0f51","_number":949483,"virtual_id_number":949483,"owner":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-08 09:48:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},{"value":-1,"date":"2026-06-08 05:51:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-12 10:28:29.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"state":"REVIEWER"},{"updated":"2025-05-12 10:28:29.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"state":"REVIEWER"},{"updated":"2025-05-12 10:28:29.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"state":"REVIEWER"},{"updated":"2025-05-12 10:39:06.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-12 11:06:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-12 11:39:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-05-12 12:13:39.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-05-12 13:10:00.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-12 14:04:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-12 15:23:26.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-05-12 16:36:28.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-12 18:14:46.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-05-13 05:25:52.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"state":"REVIEWER"},{"updated":"2025-05-13 15:48:58.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-05-14 06:50:03.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-16 06:03:34.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"CC"},{"updated":"2025-05-28 19:27:19.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-06-06 14:41:35.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-07-28 11:26:10.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-10-09 21:52:41.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-11-13 12:22:25.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-11-20 08:47:57.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2026-01-05 17:40:51.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-01-06 11:16:34.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-22 05:40:09.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-22 07:31:31.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-28 07:00:58.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-04-14 18:02:04.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-28 21:42:01.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-04-28 21:52:49.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-04-28 22:03:43.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"}],"messages":[{"id":"060520efec9f8ca3c682a19bee5e8f834e425431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-12 10:24:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8f9c5d38b7a4e056bb77a8c4f5576de59d949e7","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-12 10:28:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a610b3266c5b7ba5b56abfdb250facc82ea5d425","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-12 10:39:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3b9e010550c248b1a5ce1b48ac505bf0 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dc32d9beb818440195ca6bd284f238d0 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"f9b9e965da4267675476834024d7692eb92f7b28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 11:06:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4da44c421c3a4869ab8aeffdc59ae695\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/661a965de9ad4b87b99365e5ef6e3783 : SUCCESS in 12m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/483d9cc700854c5fb250821e671c293a : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fa34ba09ba034d25961189d33a05a5d9 : SUCCESS in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"107ad80a03c60ba5e6960972ac4382aed3fc7c70","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-12 11:39:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/835/.","accounts_in_message":[],"_revision_number":1},{"id":"01f114a563c9442a93de53790613fded370a27d4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-12 11:43:54.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/713/.","accounts_in_message":[],"_revision_number":1},{"id":"3f433c8309bb7fe80d598eec6ff4428775b1c689","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-12 12:13:39.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/345f7103e4b544499b241f5e86599678\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca167dece066479ca4796627c5c852f2 : SUCCESS in 46m 32s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/76bf483835164fc49cbf5b95ffbc6fdc : NODE_FAILURE Node request 200-0000051414 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e23e2224da0a4878aa012abaa5ac9de4 : FAILURE in 1h 44m 42s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c84e3716d48d45b1bbc7da2d33195c83 : SUCCESS in 49m 36s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca712faa44cf4e0a9f724cc32c49628d : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b06252156cd74421958e06c057b81472 : NODE_FAILURE Node request 200-0000051418 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a1d2e10520b945efbb45a7b081a7a339 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dcc314318fed45ae95c33cc781c88b18 : RETRY_LIMIT in 29s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#68\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#416\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#82\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#416\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devs... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"e54f6231c851bbe35acc21da4677b652b3e6a307","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-12 13:10:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6d08af4deb20f71ca05ac7ed4762aa2925d77128","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 14:04:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06bf2ae078454abe97635da2ccc17a1a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e581f1b344d45f0a22290a2efa29dfa : FAILURE in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f75e863575f475e9e9947d478aab41f : SUCCESS in 10m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e57934751d224e2689eb68b86b6b1732 : SUCCESS in 10m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f229972f413418db4df6b946caff6b3 : TIMED_OUT in 41m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d90897f02c146f692a78276ad4de347 : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/0e516d9586a947dfa33d9c5e98734c9d : SUCCESS in 1h 02m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1bb7938efa73483aaba7d23464e3c3c9 : SUCCESS in 34m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35123f7d1c254dabac8446c6f586dfc2 : SUCCESS in 1h 38m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5afa07c8bb341628a1f545754a243d0 : SUCCESS in 59m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/139e7ba18e904b249a16b65eb1ee4b55 : SUCCESS in 21m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/257ec0b3b99749ed9a2d8f8b2ade4193 : SUCCESS in 5m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8345ad48e3654ca59dd80a7f2ba457d0 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e13482b9d3f4b009f7ab00fd7af07ba : SUCCESS in 8m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1b9ee635a29741a0a5f7646f4eb75687 : SUCCESS in 7m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cf7ffcb951f1414285868e37a5d2b917 : FAILURE in 29m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89e534e3054240b5a03d769f38cb87aa : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1889c55e7a47448298e3337b05aaf942 : FAILURE in 1h 58m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5e8be5169d0444669860e481944f9427 : FAILURE in 2h 16m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0383302cf6334509aa0f8fcef15ab09a : FAILURE in 17m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/16db85e7d5c74fa18691ae7613cf61b9 : FAILURE in 20m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fe6a82b5a1f840b8ab1d471b6ef154f5 : SUCCESS in 27m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/66093b33b7fe4446aa255801f074e580 : SUCCESS in 1h 02m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/514bb464f2d2486681300d75c4e78095 : SUCCESS in 1h 29m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/476f75f67a3f41ba989750a09c94188d : SUCCESS in 31m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c5d347f7a10445378cef5b5aca5a4487 : TIMED_OUT in 3h 33m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a1b9accb66ce4dc2a96e452573059b0b : FAILURE in 20m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29468ede00bc4f8fabd7e27b1e1eb21a : SUCCESS in 1h 32m 51s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9514227efa4e4022b4793425b2083cb0 : SUCCESS in 1h 21m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9177d3d26294c368321660689e3e673 : SUCCESS in 53m 07s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_iscsi.py","accounts_in_message":[],"_revision_number":1},{"id":"e06088cd3f2d693821447181c7e9674851a35681","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-12 15:23:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 27m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1a5325bfda40e09ed340179fc9d58097bd6c9971","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-12 16:36:28.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"bad98c3886d56fb37cf37bd2c5994bbba1d274c3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-05-12 18:14:46.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-1/FC/ : FAILURE in 7h 49m 49s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-1/iSCSI/ : UNSTABLE in 53s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"2e1ef0e350a85561b818777a1bab96c50dd7b9f9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-13 15:48:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/1/20250513200114/NEC-V-ISCSI : SUCCESS in 3h 59m 06s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/1/20250513200114/NEC-V-FC : SUCCESS in 4h 47m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d9c659740953c4c28452cd390dc60334faeed628","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-14 06:50:03.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/1/20250514060242/NEC-ISCSI : FAILURE in 9h 33m 37s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/1/20250514060242/NEC-FC : FAILURE in 9h 47m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4d804c485b4ed18debf71555ee81f8e27060e66e","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-05-16 06:03:34.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"872727a45531fa6cec99653e1e0b11284294fa11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-20 07:25:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6af20b3513fa3827436feadad291cbd37893381d","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-05-20 07:28:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"efb9aec8efa210fd0097b7bcadc035408a97b600","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-20 07:34:24.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"acd9659d22ee3c6773e291d1b60c64447ef335c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-20 07:35:06.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"2726d3f7261e903bc2c256d79f9974440633a679","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 08:03:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60eee5fb506a4db08116c312a4929faa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a484220ef9ad445b9d17847f04067798 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c5b4316690f44611a25e0c8074efa2a6 : SUCCESS in 19m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/78c736240e3f4ade8aa2f06c7519d8fd : SUCCESS in 26m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e96488a7a2cc85a3abaea47466bafb08d86c99f4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-20 08:27:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":4},{"id":"41c0d48d9cb5ab5305ca066c50a9bc871c1ec9f5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-20 08:57:52.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"39b6474fe2b198ebd15b4ebfd69932cc02f59ab9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-20 08:58:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 19m 40s","accounts_in_message":[],"_revision_number":4},{"id":"5918c4be74fcfe38063aff1f6732529167b3aae7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-20 09:22:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"78b474ad918bcda04fb6d8b7ac9f4f2f5d0eff94","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-20 09:24:57.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"14a14c755cf30d4c6651cf09d50d4768137be75c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-20 09:57:16.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/870/.","accounts_in_message":[],"_revision_number":5},{"id":"6b5cee3276c84d71e19cb83dc9fa5044f770e24b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-20 09:58:27.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/748/.","accounts_in_message":[],"_revision_number":5},{"id":"751da33c8c7f3aa131fdcb809ad4949890bd349c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-20 10:29:35.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"3355fe31efe9f8cc753d9e936473727aa81bc115","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 10:33:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c62d1ab0428b4a7ba3edc3f7226a7288\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9a9cb119caa547d792b53e0d7c6bc04f : SUCCESS in 20m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab7338b18e414fd08bfa3e54fade385c : SUCCESS in 13m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ebfada1ccb904c3990ed0ef13873c2df : SUCCESS in 28m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9001468bec26389dee6d964027009747c80095a5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-20 10:54:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : SUCCESS in 47m 32s","accounts_in_message":[],"_revision_number":5},{"id":"af68d6b4c762e962418eae2b935df84ed2f11b9c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-20 11:13:45.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d924414279ec4ed289a093c58faa9eb3\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f574eebe6a574df0bac9bafc7f4fc709 : NODE_FAILURE Node request 200-0000053164 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e2a7323417934a5f863808ec25b0a1b2 : NODE_FAILURE Node request 200-0000053165 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7f63d3b79d4f4f599ba4d237eb77f495 : SUCCESS in 46m 18s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2afdf28cd2a14d2db6589ff060584a02 : NODE_FAILURE Node request 200-0000053167 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6519a396857f425ba1243dc96f0fce0a : FAILURE in 1h 50m 51s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a913b2d67d524617b9ccbddfaf951d67 : SUCCESS in 50m 43s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e12fe39635144a0ca8f4efe4692e340f : RETRY_LIMIT in 26s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/77ab13019c334ce582f23231c4f23c2f : SUCCESS in 49m 34s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ce23c8af010d42a0a35ae91bde72f4a7 : FAILURE in 1h 44m 37s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e26172e7b37b4f71bb96ae5c8744d183 : RETRY_LIMIT in 24s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJ... (truncated)","accounts_in_message":[],"_revision_number":5},{"id":"aab2bbe973b5a522d209a50657b45235019a66f7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-20 11:56:54.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 18m 48s","accounts_in_message":[],"_revision_number":5},{"id":"3502d252148490daa573b3e48389f4f1f2cdd695","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-20 12:12:01.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"98a30c852e89178a710c5cbcf33c4078a0e35d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 12:18:51.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a7a65f990524706ad7adc98730fff07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e1988b9deb4cdabe005c41ba40e07b : FAILURE in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f6217c4db5c4d748aa5761e42c3eef2 : SUCCESS in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7e9b092ce734974b5f288abcba448b8 : SUCCESS in 10m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e7649e15eb24626ae26f674cdff4d85 : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40b9b3ee14f4bd7a4cf2f1f511241be : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/9d84b64147c9460f8316aa997510a6d8 : SUCCESS in 58m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bc7cf86fea0d46278fc5daccd28379a3 : SUCCESS in 57m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/403373e9239444388bf0dca44a2660a8 : SUCCESS in 1h 50m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65f6acd207d44dea878daa2d4ab5c0ee : SUCCESS in 1h 02m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8276a6c0152403bb963430cdfa56372 : SUCCESS in 11m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/657ef6b183804a4f90a2fe12112fce45 : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4a4823b91e3b40c199c58d4d62e3479e : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a4401bf1c8054188bcf0f8e41d10a205 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fa64266a726f45f984b47242ea7e640c : SUCCESS in 3m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/71dd901cb6af4e11ad498540723665d7 : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ecc4ef3c559f4b0da21f1cf89c31b485 : FAILURE in 33m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01ed2b1f84cd4eb2bc41b62809ff2743 : SUCCESS in 7m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/466528500a804c0b94f22f9388214bcd : SUCCESS in 1h 56m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/569f6307b423425a8a4ba54a2d28121f : FAILURE in 2h 25m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/91acf71b8b554911a5c1d4aff39f11fd : SUCCESS in 1h 58m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/601e2707c78e4d5588578426eb6589eb : FAILURE in 25m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/94c390fbd53b4818b3964d45a5dd4b76 : SUCCESS in 17m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d99ad40b6ef74648a62acf9baabcfa0d : SUCCESS in 1h 08m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9375cfdb9c1346d7b0c120b70ced6824 : SUCCESS in 1h 31m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3d40ec48611a4f0ebb43fdce39fe9ffb : SUCCESS in 37m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1cdee37e27d84f1684dfb5e4cdfb6618 : SUCCESS in 1h 19m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d7e075e09b584df393fd80d7d1595ae0 : FAILURE in 25m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16856cc2aeb54865865d3409998892ae : SUCCESS in 1h 32m 38s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9df7a6aa0cd645c9976119bc8a1652ee : SUCCESS in 1h 45m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd17839e14de4d4e885a09932b07e6c1 : SUCCESS in 31m 50s","accounts_in_message":[],"_revision_number":5},{"id":"da778286f3245f15625b36051d525fc12928d450","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-20 13:06:45.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"c4fb35d6ed2cf484e1ea8f20708fde79a67bc164","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-05-20 14:06:39.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-2/FC/ : FAILURE in 6h 40m 47s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-2/iSCSI/ : SUCCESS in 1h 48m 59s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"7828eb4cc303d74e7c1a749a6433870340e85f7a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-20 14:29:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/20673f15d4eb47ee881bc775237ac62e : SUCCESS in 1h 17m 59s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7757180bb70443a7b995ba41c70fe119 : SUCCESS in 1h 19m 52s","accounts_in_message":[],"_revision_number":5},{"id":"1e9a67abb3fd472e231a9dfb68696e8c4182b292","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-05-20 15:35:56.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-3/FC/ : FAILURE in 8h 1m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-3/iSCSI/ : FAILURE in 3h 16m 16s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"97a88dc0bac456dbcb99f65366854cd7f3829e0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-20 15:39:22.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"fec52725e08a01064c0264e12e9f3b285afebe0d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-20 16:10:50.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/756/.","accounts_in_message":[],"_revision_number":6},{"id":"0e9927411f71a32f31b074e5288ff26a02088c4d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-20 16:16:18.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/878/.","accounts_in_message":[],"_revision_number":6},{"id":"edf0087a45507547d8d4bfa423e67e1bd99cc4b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 16:51:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/372eeab8383b4c4489f7fe7daa588298\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/37d7662f7ed74f41a46d15856275832c : SUCCESS in 19m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a080cfcfe34a457985ed02f87c4e3c04 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fb54c8f603154d7aa9a77d989e105062 : SUCCESS in 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"441429df796261dfe7e595cd08f72efa1ed0677e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-20 17:28:02.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4ebc82a553754d5bbaddddb53f8531de\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7617bbfe36e7415eb56d5ff663d40e65 : NODE_FAILURE Node request 200-0000053721 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bfbd347123e547458f55b42317e60026 : NODE_FAILURE Node request 200-0000053722 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e4556b663c2b4dfcbfd7fe309f00f696 : SUCCESS in 40m 47s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd85470cb9e64af1a071d6b8eb7a2f70 : SUCCESS in 46m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3e475214191e4a5abd2eca19fb6fee17 : FAILURE in 1h 38m 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6705af5e2e9e400a900827606fd863f6 : SUCCESS in 46m 42s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9da02df789b0445188dc9b83fce79a41 : FAILURE in 57m 41s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/46288abe40754bb0957ff4d8042b3433 : SUCCESS in 50m 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d4efc8ffda744754b704e3aa59e848e9 : FAILURE in 1h 48m 07s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ff6c5028fb344e62b53974eff06450d4 : SUCCESS in 47m 22s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None ... (truncated)","accounts_in_message":[],"_revision_number":6},{"id":"1bd2e7cce889507288adccadac997bfc06294c41","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-20 17:28:55.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":6},{"id":"1f2ed6c858ff4757ef28628d9307701287757957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:16:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45402ea706d74cb8af3020f90ef98abb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7933e49c54204d63b36dd98ae6ddbae2 : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c89fdc86ad384aa78d408ee38e7840e0 : SUCCESS in 10m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df4fd85e80fe478baab59d465ac481c6 : SUCCESS in 7m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0eb5e4e83c4a4a4da2f773292e620bde : SUCCESS in 15m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65bdb3985b33457f89db230f8677c092 : SUCCESS in 6m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/b2ea18514fdc47b7b25efb9a59e6fa95 : SUCCESS in 52m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/be44bd0d87c64247814df10ae9cfaa61 : SUCCESS in 58m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28a3b850a2b34378a26798d27149caac : SUCCESS in 1h 25m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72133d0b1afc465183c7d4e1222a449e : SUCCESS in 1h 07m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cd133e3390247aca507daecf3dc4fc1 : SUCCESS in 19m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce25b23392a44f38b890ca2be9b4142b : SUCCESS in 27m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/422e4497bf7b4f569752cc7dcae83e7e : SUCCESS in 5m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e937610559f243409fee6ca10e3b4e4d : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42fc989e60da4422ab124c65c265249f : SUCCESS in 6m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/be7a959abe53493b8b0e688921110507 : SUCCESS in 8m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc9ae6ba04c8481f8cfc4c8ab36c8838 : FAILURE in 34m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ec630a9800b43a4bbcb6f95081d7361 : SUCCESS in 7m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/53cfd06b9b4b4e768048769551016a9c : SUCCESS in 2h 10m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/150d5ea051df4f1d8a63badc8bd3e15c : FAILURE in 2h 31m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/40b7d24456d541b0887ba662397f7db2 : SUCCESS in 1h 53m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/87e7d7c7fb6340f4b7a60ba764ec40c4 : FAILURE in 26m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/218dd0023d3f4b00a875ca927150f663 : SUCCESS in 33m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a33b8f929a5642a6a68bb14385d9cf18 : SUCCESS in 1h 10m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/db26b1aa1ee8467c93d0540c77f4538f : FAILURE in 1h 56m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ac61e84c7d6d43e582f9f3eb18cdc3e0 : SUCCESS in 39m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fdddffc9a4ad480987ce92ffef4a9ddf : POST_FAILURE in 2h 11m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/53f0ec8bdb654c19964fef973d34e09b : FAILURE in 16m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9c4b6edd81dd4715a8d45bdc6018187c : SUCCESS in 47m 55s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a04bf8bf3608432490adcb7a3acf9b50 : SUCCESS in 1h 25m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28214f313c624b17a101f2ee38494332 : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":6},{"id":"c295824437302c5b80487ae8b392006367c858f7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-20 19:35:10.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"316d678264a46f45d469e42466a7b32822e8df83","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-20 20:35:43.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 20m 33s","accounts_in_message":[],"_revision_number":6},{"id":"05e040071d71b0f1d5404e9389c1993853a03212","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-05-20 22:49:15.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-6/FC/ : FAILURE in 7h 9m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-6/iSCSI/ : UNSTABLE in 24s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"746825ba810b10ea0e90d1f959b2ee208259ff8f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-21 00:26:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f02f7098fe334a89af0210ab647d8cee : SUCCESS in 1h 17m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/538141d356524c45b2e968875cc23ab5 : RETRY_LIMIT in 2h 48m 07s","accounts_in_message":[],"_revision_number":6},{"id":"c9f68032aad19d29b6823eb200438f34b62ac765","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-21 17:19:39.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/6/20250521214858/NEC-V-ISCSI : SUCCESS in 4h 19m 14s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/6/20250521214858/NEC-V-FC : SUCCESS in 4h 30m 39s","accounts_in_message":[],"_revision_number":6},{"id":"75ccac92f81a869968fedd7b2e82f1f2e4a9b7e2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-21 22:18:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/6/20250522010832/NEC-ISCSI : SUCCESS in 5h 42m 30s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/6/20250522010832/NEC-FC : SUCCESS in 6h 09m 45s","accounts_in_message":[],"_revision_number":6},{"id":"af0fbbf26b4f0fe0fb2ac8155601ebdd92d2b59d","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-22 08:56:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"849427e9fc917f210396a7015317b2ad8639cd2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 09:37:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f6841fa560a441d9bdb9eec83b3ecea\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2489f1deda774915b0c4f63ca1d54b87 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f8b86a3900b046ce8035b0cc2b69a25f : SUCCESS in 24m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/800511019860459faeb38a01dbdb7d86 : SUCCESS in 32m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"563dc853af85217fd58c143fa7fca5db96dafcd5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-22 10:19:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/99152ec649b7408085c156ed994953b5 : SUCCESS in 1h 21m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ccdb07ba4a234c99af9feaf06a5c9fe9 : SUCCESS in 1h 17m 02s","accounts_in_message":[],"_revision_number":6},{"id":"e18c58dabb10e110106da60c2c1b56b6d6401655","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-22 10:33:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/720863483a5245428951e057bdff9e98\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/686569d8b87a49a396e954a3c779daa3 : NODE_FAILURE Node request 200-0000054036 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/963932b03b4e4752989ab1dc2dd6640a : NODE_FAILURE Node request 200-0000054037 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cc56206c421349f28d6a68350569b38c : SUCCESS in 51m 21s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/953f2615ce8548b5b991dc24837d779b : SUCCESS in 45m 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3813a47e0f3649f9ab13b901d7adfff3 : FAILURE in 1h 36m 48s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1809c9da52a54765b44b476e1be55c9a : SUCCESS in 50m 00s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3dd1e84e21844325ac34390ea5dea5bd : SUCCESS in 45m 26s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9d1b438ba1b44813bf7b5e91fb82f776 : SUCCESS in 47m 06s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7d7b24b130044d84b414460e4a2fcf05 : FAILURE in 1h 31m 35s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0b3f5c6e9eb64d4ba03472fa3de5ee64 : SUCCESS in 53m 54s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: No... (truncated)","accounts_in_message":[],"_revision_number":6},{"id":"2a86cdbefe23e202ac06f2bbb4a5c0ae035d37cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 11:40:21.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a07bfa3cfcc433897e7b0f2d871099a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05cd2cd64b684c958b5e8fac5a960dd4 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28fa770690114f2aa1729c9657dd1e84 : SUCCESS in 11m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71a77e058d8c46d5830d38d88582866a : SUCCESS in 12m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4e26a3cc31e45ebaed3b675633be2f7 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f1ce98663f4be7927374b80484e6f6 : SUCCESS in 16m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/8fac5f806d1c4055b0843001954b75a2 : SUCCESS in 58m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/98dacf4136b94b1f8da074aa251f8407 : SUCCESS in 1h 03m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/044c29f93e3e449abebe801f5eb24589 : SUCCESS in 1h 39m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/093d3c363f27497193e3cb77d6147bf4 : SUCCESS in 54m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea71c9eb313a455482719baaa05af3df : SUCCESS in 16m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4e0d7e5d70d64eb58378bffdfc108acf : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/46d12bd92e304c61af717f83873993bd : SUCCESS in 6m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76e22e0daff74254ba3f48ff160e9b2e : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6f9823b0f45a410c8bf371e6bf36f368 : SUCCESS in 7m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b9cc6583a58d45b4b1cb97ccf0e07534 : SUCCESS in 8m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3d4273d6b0bf4da1800c493609b9c4a6 : FAILURE in 33m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/223dd3ea30d94b23b330501e9e6a1259 : SUCCESS in 7m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ae5db39143c9420ca2bb6f6d564c2210 : SUCCESS in 1h 57m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/72f63f42d46b43da96013553603ed158 : FAILURE in 2h 24m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a027b89c420d479e9739de734beccf1c : SUCCESS in 1h 37m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ce159137472c4cfa9ea8166308c75244 : FAILURE in 21m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ca8bb7727d5b47c98c131b39a2dd1f7d : SUCCESS in 33m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cd6c890601e74d15a379daac7b8970fc : SUCCESS in 1h 00m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2e60c25e064a42c4bbc1d539bf1d2d7d : SUCCESS in 1h 32m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/92f027555ca14216a2e57c4c1d38cbff : SUCCESS in 42m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/038a1d1234d54ae087bf7d5b6999731e : POST_FAILURE in 1h 42m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1b13aea6e5744ea59b3f6723181194cd : FAILURE in 12m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9b9e4d2a59a4d6cb4b137e1ab4a4ad6 : SUCCESS in 1h 45m 10s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5be6f0e5625945c5a6bef09cd1f8b176 : SUCCESS in 1h 54m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9207412460be450992b6aa6ef28fb3a6 : SUCCESS in 56m 08s","accounts_in_message":[],"_revision_number":6},{"id":"236ea8604f40e65e59e05a300030490bfe16bde0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-05-26 12:14:54.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"e87cd964ee6eca8a329dac81cad705ae31946d9a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-26 12:50:45.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/db66e215008147f3967a987033266b36\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/40cc1f27afed48f392846ed6912e3e17 : NODE_FAILURE Node request 200-0000054468 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8ae7d642c88046a49c2cd64d6009e886 : NODE_FAILURE Node request 200-0000054469 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ad2e093c61be4c56879fd3ed8ec226b0 : FAILURE in 25m 59s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/413f3626ebcb4af19fc7cb105042c839 : FAILURE in 24m 49s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5a73fd500b644a978d8e364df7d2c12e : FAILURE in 26m 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ba44f188b314ef4a57dcc7a85732c06 : FAILURE in 25m 24s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3a12fb9a8bf54b66a93d65e886204cd3 : FAILURE in 28m 15s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/60e362c8dcdc4a5fbb2f04144920f72a : FAILURE in 29m 44s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/147051e0b6b444f38ec76ef1b2d41e3e : FAILURE in 28m 53s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/37b3dbc0ee104828aa8c7e7c9abdaf3e : FAILURE in 34m 42s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source... (truncated)","accounts_in_message":[],"_revision_number":7},{"id":"ab59c7b7c5eafcc20fd78ba1fc2468427cc7a356","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 12:50:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d946aceac5134f20849731584cff2312\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b0269fb510a4358a0b5228ba42ff4e6 : SUCCESS in 16m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1de2d439d4c04861a5f2f4ad713935c6 : SUCCESS in 12m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/84b326215cdb412a8ebb78bec75438cb : SUCCESS in 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4041e7af31c25a3bf07dea123a033191bd35254e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-26 13:36:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/380f4189412e42d099767b4c2526180e : SUCCESS in 1h 12m 43s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c3abcca7f6924d358494cba9ec1d4ba7 : SUCCESS in 1h 19m 39s","accounts_in_message":[],"_revision_number":7},{"id":"ba89cb76725b59c2bad5a93d25b1f5529ac52384","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-26 13:56:26.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":7},{"id":"0fd317cdb29163b550d59983304ba1bb2e2bb796","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-05-26 14:15:10.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-7/FC/ : FAILURE in 1h 59m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-7/iSCSI/ : FAILURE in 1h 56m 12s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"666178640070084f4365ad634e99c11b97ec47fb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-26 14:34:04.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/909/.","accounts_in_message":[],"_revision_number":7},{"id":"ed1d9c8c98ab69df131ae33de19f44f8dea63137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 14:40:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d878b52685a3415593b895ea8d9460de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c4c7c8e09f24c8893350fc777633a1c : FAILURE in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8897de690fac4af5b5f21963ad539069 : SUCCESS in 11m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c88a1a816e30455682b466feacd4b3b7 : SUCCESS in 13m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9fb986652004f888dbf51a0098fefff : SUCCESS in 14m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f63c107ca1c40ee86abcd42b96034be : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/1a9dc2c10cfc4639b7c5a7eed2866412 : SUCCESS in 55m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/19be3c530af44e5d967b5505c1753d74 : SUCCESS in 1h 04m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/642d4029fea14f069242354a56e4da68 : SUCCESS in 1h 40m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/327667ff40ff46e49bd8b0919f481524 : SUCCESS in 56m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef5f10d6acd741558a0aa9211c5de4b8 : SUCCESS in 20m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d802c68865844785ad9d9ddf3bf00786 : SUCCESS in 25m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c73db6335f2f44f08a4381c9f9662fe4 : SUCCESS in 5m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e82d1519acca437eb2434c183b3bb685 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/19c9a5fda6d648b386200e368c39aab7 : SUCCESS in 6m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cbef405643204c17a5137a9c5272d96a : SUCCESS in 7m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/278d28cc0d484e22a27cc96dd0ce28cd : FAILURE in 29m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a22156dc20f64606942c43e2de84cfbc : SUCCESS in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2aacecd4683240f7beff0c548b9bb8b4 : SUCCESS in 1h 49m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e4122aa41e584076a0bdd575cf1f01da : FAILURE in 2h 12m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5bb071f5f5d048c4923528965f6156fd : SUCCESS in 1h 52m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/55e6173eafc64dbb8dfbb027f0eead53 : FAILURE in 18m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f0214ba48ea4489ca71c9fb94ad9b1e8 : SUCCESS in 29m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0eb0b3b0f66743ada56bdc46b49d973b : SUCCESS in 1h 08m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/25df2747a3f4443d9f60aa80e4394b0b : SUCCESS in 1h 49m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/35ba0a0e3a9044f98adf58a57ee68097 : SUCCESS in 39m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6b51ed69be004a749636b13aff283205 : SUCCESS in 1h 38m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/432c9c1276a649f4933c4c8152140e16 : FAILURE in 23m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b860f2019cae47d7924e556808b273d6 : SUCCESS in 1h 36m 34s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4ba293f067e4aa19780cd5090634e55 : SUCCESS in 1h 37m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd7d579cafbf41fc8582df3effa49651 : SUCCESS in 1h 00m 59s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_iscsi.py","accounts_in_message":[],"_revision_number":7},{"id":"e0a6fde6103c4f6cdbee9e8a980308f75da7b39a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-26 15:02:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 17m 45s","accounts_in_message":[],"_revision_number":7},{"id":"0907b5c2a2e53eb8bbb4331d3a1c1a6b2a0512d3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-26 15:02:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/783/.","accounts_in_message":[],"_revision_number":7},{"id":"430c6a7528c2c236125b6d8e091d699589751cc8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-26 16:43:44.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"e578a94ab3df7ee312cc9f2f19c6439d6c5de844","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-26 21:32:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/7/20250527015820/NEC-V-FC : SUCCESS in 4h 10m 11s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/7/20250527015820/NEC-V-ISCSI : SUCCESS in 4h 34m 35s","accounts_in_message":[],"_revision_number":7},{"id":"a76fc0e196e94130de92b4452c123b1801c0658b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-27 08:11:49.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/7/20250527082116/NEC-FC : SUCCESS in 6h 10m 33s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/7/20250527082116/NEC-ISCSI : SUCCESS in 8h 50m 30s","accounts_in_message":[],"_revision_number":7},{"id":"d01fef696facc394e5b9bd404b7eef858222f12b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-05-28 19:27:19.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a6d05e158bdfc74deaee6c67d5f2fbbb6207516c","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-29 05:46:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a3f383c8d8bdae835df7fb34dba525047b7379fb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-29 08:12:14.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b3fe1b9912d741cca7bdf4d5f9c4efd8 : SUCCESS in 1h 44m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/65adea2ce8754c1d856458bc29bbe186 : RETRY_LIMIT in 20m 06s","accounts_in_message":[],"_revision_number":7},{"id":"8a5984daaa874ad3b5f754339976eb7824719db1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"real_author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-30 06:53:29.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_36800\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"}],"_revision_number":8},{"id":"342898c5309e0ef64408c7a85a3bbbad30e3d58f","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-30 07:06:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4d0d353954995cddb789402815213d6203cf7470","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-05-30 07:07:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"593387707ebfb6056f1e0ff39ac7027926b2555f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-30 07:08:05.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/925/.","accounts_in_message":[],"_revision_number":8},{"id":"8897024ee4f4abfa613680eb4929c93343e7e017","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-30 07:42:18.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":8},{"id":"5930e0e457d6dcba7176626c215f94f413e044a9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-30 07:49:37.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/801/.","accounts_in_message":[],"_revision_number":8},{"id":"72fc46db0f038447c59d1cf086920f8f05110b3d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-30 08:14:02.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 18m 48s","accounts_in_message":[],"_revision_number":8},{"id":"3167aa5ebada28083f9ef1415ae8ad89bf40b0dd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-30 08:14:47.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e6cf0f55446740d4b7282f161567a786 : SUCCESS in 1h 15m 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9f548f8f13f84e95a6f2354d7be2d9c7 : SUCCESS in 1h 19m 24s","accounts_in_message":[],"_revision_number":8},{"id":"3b6ae89e5adeb5bc36efae769eaab2aae5029332","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-30 08:23:39.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"968ff0b9b1f0853c4ec6b766b16c4b719117b279","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-01 04:38:17.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/8/20250601085619/NEC-V-FC : SUCCESS in 4h 38m 36s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/8/20250601085619/NEC-V-ISCSI : SUCCESS in 4h 41m 56s","accounts_in_message":[],"_revision_number":8},{"id":"6a1b5961be38cc3ecd9743dd935bcdf272968d44","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-01 08:43:18.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/8/20250601113144/NEC-ISCSI : SUCCESS in 5h 46m 32s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/8/20250601113144/NEC-FC : SUCCESS in 6h 11m 30s","accounts_in_message":[],"_revision_number":8},{"id":"59613a67f6b5354da4bf86400d604d91ef08a6aa","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-06-02 06:28:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8f22576f19d545ce5b7d52cb5f7d778e05ff35ce","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-02 08:10:15.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/09cba47110664c93bae06840ef948fb8 : SUCCESS in 1h 24m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e6456689072b40529a1667f2ad89f961 : SUCCESS in 1h 33m 56s","accounts_in_message":[],"_revision_number":8},{"id":"42d039d0edb8219c5d41e367368ab94331cf31ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"real_author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-05 12:08:59.000000000","message":"Patch Set 9: Patch Set 8 was rebased on behalf of \u003cGERRIT_ACCOUNT_36800\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"}],"_revision_number":9},{"id":"8e0c5a83048a71190e476b85b86c2d8f0b66909d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-05 12:17:59.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent tree, and commit message as Patch Set 9.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":10},{"id":"3d725f3ccdd3c58764d03759117e52d15b2dad14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 13:06:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88b9ee0068fb4ed1a93b20674642f7be\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b6fe6573979f4ad5824692edfde16b48 : FAILURE in 30m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3cd683076a54261b1a1d9b110a9660b : FAILURE in 29m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2efa91a09ac44d9d869a47f29eac011b : FAILURE in 20m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"619a6338dfd418e29b5e325e3b599ce3e647936c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 14:32:03.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ee755f7eeb34719b3fda1757872fc1c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77089084fa76403582cd40a40ad5e2d4 : FAILURE in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/075513fdda274226b522199f1cbcddac : FAILURE in 10m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7e299207eb14de9818caa313afb97d7 : FAILURE in 11m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f5dc83d527d43949a0256c9a3253221 : FAILURE in 15m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cad0e5e62a7f42a6a03e0c90dd1b4be2 : SUCCESS in 17m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/a8192e1376e44df1b500511ab53c1abb : SUCCESS in 57m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee1312bf84ee4652bf19232d4db40f2b : SUCCESS in 1h 10m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e30b0d0fae99426fb5c88c868214338a : SUCCESS in 1h 37m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/039e4345882f42e191adda6747671f66 : SUCCESS in 43m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/513dc9b61f6d4e7da3498c07ecc385bf : SUCCESS in 15m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7e4e3cbdee314ba0a93e1d2127e8e4cd : FAILURE in 21m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c1b3ad8b253f41999bee9627a79a25aa : SUCCESS in 6m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7200be55713943588c4688d8aa9846e3 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/139e71ce523c4744980f3367b78fa991 : SUCCESS in 6m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8286c1c616d4497ea27fcea0781d7e07 : SUCCESS in 9m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c019f99eb23e4438a28b0aa8c9dffc7b : FAILURE in 39m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cbc43d01d83943be921f9ceaa530a4b6 : SUCCESS in 7m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f63e7641dc34f1daf35a017218079cd : SUCCESS in 1h 54m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/40256a4b42ff4c14871ed07d23d29ba5 : FAILURE in 1h 32m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e551e67d32e64556b399de56ea2d2ede : SUCCESS in 1h 46m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9288f2c746fb4d0eaf0711970b03ab68 : FAILURE in 19m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/569e42a36abe48adb76573ffb4381a04 : SUCCESS in 30m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0ba19d9dd7254ba3990f281df7ed85ad : SUCCESS in 1h 08m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/77610a54b55d429296782a214f8885d8 : SUCCESS in 1h 39m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/79564e380ca04a28a126dfa1ec74dd9a : SUCCESS in 40m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/845c18d0681145978607c4e12f085717 : SUCCESS in 1h 30m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9573733cb197452193914ff8b89bb483 : FAILURE in 20m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2ee287ef7bc41ab84926f2a5adcb894 : SUCCESS in 2h 01m 21s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4d95d6247034bffbf337b6a24b49cd8 : SUCCESS in 1h 38m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3551f0fa6139445493afd6937813e6cd : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":10},{"id":"eabf3895069043068767a6b45fc6b4fb3116dd32","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-05 18:58:56.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"ad2dc579384173db4f5d1e90a08fbf2e546132e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-05 19:15:24.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"6b6e095ffd7a1c499df79d2fc218c1fe8343ebbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 19:49:48.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ec3a71affb7426bb10ba47ac62d180d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d8f2000e4f6240e5905197d551046647 : FAILURE in 30m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2db739efbd1e4e0ea235bc848c9dfb33 : FAILURE in 31m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/df90716eff8041be98d2ff95afbe6583 : FAILURE in 31m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9a7fc13662f4dbb46c1cb2bc1e272a02d093decb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-05 20:18:03.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"189424c9defc45250b6b61a17279d2578b12293b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-05 20:23:06.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9d0e227f298c42179634d2930d547ea5\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a568b6489cf4fc295c1ba117b6aa729 : NODE_FAILURE Node request 200-0000061790 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ec2ad29d83ae42c2aa037b8c2f2156a7 : NODE_FAILURE Node request 200-0000061791 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8043314cc99d4b0592d4ea1c308e18b5 : SUCCESS in 50m 40s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/80e5fa4f77044219b3736f4a9246c172 : NODE_FAILURE Node request 200-0000061793 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/66f2df47259248e5bfbc13b7a78a4ba4 : SUCCESS in 49m 15s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/66f2da7d458e43db946e8252d92ed9ba : RETRY_LIMIT in 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc58e4b8c2624a28b42a2b65175e28b1 : NODE_FAILURE Node request 200-0000061796 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dfcebce4aaec4efcb19906992f058221 : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":11},{"id":"cc07a8614a50b28496f7a222c1a808d069481f81","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-05 21:22:36.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-11/FC/ : UNSTABLE in 2m 18s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-11/iSCSI/ : FAILURE in 2h 6m 50s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"08721976aba5c1d85f27f27d6c35bcecf076f597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 22:13:01.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/276cc0c6aa6a4d17852868d478aa8ccb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a162a8448d414f02876188d6ea0f0d18 : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/150d2f109f6645be83df6ceb94be8532 : FAILURE in 9m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01da1a4ef8f3464192dc6f961feb2df6 : FAILURE in 10m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2638bddd97bb41b9b6e1367168e3e494 : FAILURE in 14m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5d476f41df84f2aa2cb5609125de8a5 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/756b814bf54e4bbeac5904a440083b8f : SUCCESS in 54m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c1ddd6b48544844913f621a4398e7dc : FAILURE in 27m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4e56b4a75df3429a88dee2daf47770b1 : SUCCESS in 1h 23m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/163b611ef023439589de9648b9222d78 : SUCCESS in 1h 03m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5aef02df0bf4fd18a6d2a2f79b34de9 : SUCCESS in 19m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/495cfa00bf1d4175bd6fb21bc86c7cce : FAILURE in 20m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/21faaad90b67470f8d8b92e206e2201e : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b96295a9074149bd9cce1a61b7bf84ce : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/49dc7a879b134cbd87803ede1e193cc6 : SUCCESS in 5m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/87702516c25b48ef81a1f362adb58602 : SUCCESS in 12m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8dcb66d885fa412bb9db074fa9b3702b : FAILURE in 32m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/92c33e40c31f46d287e42897b3e13e0e : SUCCESS in 3m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4b3119e763f94ea48ded026b1bded488 : SUCCESS in 1h 52m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5500122bb3b348c48ac04a098a8f75cc : SUCCESS in 2h 29m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02768c40463f4ca5bf1f3243677c8a5a : SUCCESS in 2h 01m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/05d47cf43c834dbabbffca1e46ce0ef5 : FAILURE in 24m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c9fd5e5789f5421985fac330eec5f97f : SUCCESS in 33m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d498f7de10bf4c9eb100316b1853f587 : SUCCESS in 1h 08m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5b8910cf91284080a0afbf88c7df67e1 : SUCCESS in 1h 46m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fe0ca97a0f2e4a20a2e144e3ea764546 : SUCCESS in 54m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dd363af3ebe54e1f8885433221a80581 : POST_FAILURE in 2h 56m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/622e4058003745bc947d1d33efb7463c : FAILURE in 23m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/267f6af0929347de86eace8913aa8b7d : SUCCESS in 1h 37m 28s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02be25f79acb4fb6a8b206c0e031ec93 : SUCCESS in 1h 31m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab60a1256f58457ba66adc4dd5614e61 : SUCCESS in 54m 32s","accounts_in_message":[],"_revision_number":11},{"id":"5061570349a5bf0495752015afedc656bd834c4d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-06 00:18:09.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : SUCCESS in 47m 31s","accounts_in_message":[],"_revision_number":11},{"id":"7d743fc9a0afc5259f317a05bee5f184307464c0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-06 01:19:39.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 07m 00s","accounts_in_message":[],"_revision_number":11},{"id":"07a9cab5bcf3104c64c0215c28d3da71a291b4ad","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-06 01:54:11.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"9f21220a02a44a5162ab574e6bfefde438143cee","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-06 05:04:29.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/79441aa532724edd8963faaa35b0f50f : SUCCESS in 1h 20m 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3b134aadd7494cd18c4b86099b47b07b : SUCCESS in 1h 17m 37s","accounts_in_message":[],"_revision_number":11},{"id":"3c43642b47504e489987beb28caa3f6291a2055d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-06 06:33:50.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"c078d5a4609e399d875bbf2d70f1714d47abcf2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 07:04:00.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dfdac38552d41cd96adfd2701a42890\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d4b9ff374841428ba10de6700e9a06ce : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69b1c41aa7d249f4915ffdbaa9a86915 : SUCCESS in 22m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eaf32a278c8a4e16b56f3b660616e17a : SUCCESS in 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2f42a5fe03b4f426260912c3022b624bcc476d1a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 07:26:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/875/.","accounts_in_message":[],"_revision_number":9},{"id":"6c63a120b58b25ba77aea3204e102455677ed641","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-06 07:35:50.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/61a741cc7c5b434aa9fba586c52d6672\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2068a5442e774787a0ec90246331abef : NODE_FAILURE Node request 200-0000062965 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/24032eb592b24536aaf9796f157df891 : NODE_FAILURE Node request 200-0000062966 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a69451df0b29497d84da0625c573fcf9 : SUCCESS in 42m 05s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a2f88716cd24d9895cbabb39124f6af : SUCCESS in 47m 44s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a304f65f08254883a9f073ffe56a4e66 : SUCCESS in 43m 30s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5dc5b9e0cfd145588d669527e6981fd1 : RETRY_LIMIT in 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9c0cd79bafd84d8399e3002d93cce3e8 : NODE_FAILURE Node request 200-0000062971 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1413073691e242c198665a03b8703733 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":12},{"id":"07303850f098962da712be7e8bb3cb0e58709ebb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-06 08:07:59.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"aae1b152d82c101b99c4544a07f5d888265f6b8d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 08:16:57.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/999/.","accounts_in_message":[],"_revision_number":9},{"id":"317a0650dfe71b39a7038898c6ed9635424cf786","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-06 08:19:39.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : SUCCESS in 49m 29s","accounts_in_message":[],"_revision_number":12},{"id":"a41f6aef3b29b4095c3ac31d425acfcbbda7fd0c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 08:21:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/876/.","accounts_in_message":[],"_revision_number":10},{"id":"fbf7738f8b2cdb2a3bbaf3dcfac18c7cac4fc15d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-06 08:49:45.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-12/FC/ : UNSTABLE in 2m 27s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-12/iSCSI/ : FAILURE in 2h 15m 32s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"293e0b6d48d7df3dfb98ca6dc00db011d1cf282f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-06 08:59:07.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 09m 27s","accounts_in_message":[],"_revision_number":12},{"id":"cf5f8ff474d11ed5993140e38933ce5c1fd5f83f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 09:12:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1000/.","accounts_in_message":[],"_revision_number":10},{"id":"ce3e93fa1c91bf6a2646b513229bc411d483be50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 09:36:55.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1360ff792f434eb8906913490864cdc3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5609b8fa2206406ca67746cb842626ef : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc9dec7470d84b3092198ad99d724c3b : SUCCESS in 10m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6d5f84500bb41189c75ff05b4138ef1 : SUCCESS in 8m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab63cf1c76604a7d87fe8735ba7696a4 : SUCCESS in 13m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1643ee283643418b9aafe2c4fcd79171 : SUCCESS in 21m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/3a338174b2b144ff98b399aa0a6ddb17 : SUCCESS in 34m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9e764c9958414be08d21822087de836d : TIMED_OUT in 3h 01m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7a96eeff52b41d5a501fe3fec715947 : SUCCESS in 1h 26m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aada212f324a4b779fdb705265cb6d84 : SUCCESS in 54m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2548c12a014247a387f3646d0df04983 : SUCCESS in 11m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/184338bc597b4c01968b0c997c83ad43 : SUCCESS in 16m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8c656476a4074201b82c6e6c44653b1d : SUCCESS in 7m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b711bf51ec3f417abdd8da94641cb980 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7312cdd134b447ed91f56ad370170ab0 : SUCCESS in 6m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/afaf4015bcde45d0a5517d2fdaa893b8 : SUCCESS in 8m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a7f69bd506734628b9985fbb426b29f8 : FAILURE in 38m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e4e0cc7bf37483a809030a6f3abdade : SUCCESS in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/32ab8b7b5aa54ec9b3e457c4c24ea879 : SUCCESS in 1h 45m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ea03c9bfaba542e194c4c2c55cd79f1d : FAILURE in 2h 47m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7fcc1715355a4df3bae72b6fce4e17e6 : SUCCESS in 1h 51m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cd989ac6cd96456dad57fe82b8396c32 : FAILURE in 24m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9cd656ed12aa4e55bd51d37e0f516051 : SUCCESS in 33m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/58732e8c9c6f48c4a346132f8ac82a24 : SUCCESS in 1h 11m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3e15f6404fdc4b0c86de5081ce9177f0 : SUCCESS in 1h 43m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2723d8e4b2bb4832ab3748b924faa5fa : SUCCESS in 38m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/22f03c7da9c7459ea613bc9a476a5f78 : POST_FAILURE in 1h 32m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e433dae5e84e498a8c7b778ad8a6f0e8 : FAILURE in 15m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66fac0944f434ecb91e682ba33448958 : SUCCESS in 1h 31m 38s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9897ffc5e69942dd9562eb031aa82ba4 : SUCCESS in 1h 32m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c51e9b82182a4ef6ad6cf500be681dda : SUCCESS in 29m 09s","accounts_in_message":[],"_revision_number":12},{"id":"26bf936a3057d69268bc674b225f15e99ebd91e4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-06 11:01:01.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/48b054285288451aa3db80ffb918ad31 : SUCCESS in 1h 13m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/590216a588c44c8abbedc862d95377b2 : SUCCESS in 1h 33m 39s","accounts_in_message":[],"_revision_number":12},{"id":"fee4ae7f964f5660c4adf6d39251182cd5bd9767","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-06 12:33:56.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a2e35aa7e2967c5899d79187d2d10949832f4179","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 12:50:26.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bb01955c8fb4994a05879c57247abf8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/73c83000ec9c4d3caf5192afe535c00e : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/20ffd24905e34ccaae37e9171d4fcc92 : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fad1bcd107314ce9951dbc88393eaa68 : SUCCESS in 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d4df0615c1ba6ff02926a590bf8e90f48657c80a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-06 13:33:40.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/681c053ff1fc4fc18b7a2f945856cebb\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ad5dc5920f674762aa827c9f9da5228c : NODE_FAILURE Node request 200-0000064667 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81f7ef761278474f8bc6f392db09ca09 : NODE_FAILURE Node request 200-0000064668 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/422016b0b76947c0a0318c5e52f8f4e5 : SUCCESS in 43m 25s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/84a7581454304c98a612954a0fe2dda4 : SUCCESS in 49m 57s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5c5ad64c200c440ab3f1f2a0feb03ca7 : SUCCESS in 49m 49s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2de54ede27c94a388dee212d69e20689 : SUCCESS in 58m 52s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/936c9b208bb14e649d40cea513dd1a58 : SUCCESS in 49m 48s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/76cbe3ccec4c4495864985a4bf800b08 : FAILURE in 55m 48s","accounts_in_message":[],"_revision_number":12},{"id":"d2e6a16b9dfbc4b96f1303f7e90ed829d56cca6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 14:41:14.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/943364 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"0d9ce10335892d7d5a9a43a0116f42a6272cbd01","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-06 14:41:25.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"e3b73e30e1751e43952fe487031cd02b73239f66","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-06-06 14:41:35.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"883854cc410951888625d8eb3ddda8c3172989f4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 17:29:15.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1009/.","accounts_in_message":[],"_revision_number":11},{"id":"0ce9f178040d8f11f5d1465ffb21ba63ca944a6e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 17:55:26.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/885/.","accounts_in_message":[],"_revision_number":11},{"id":"78a0d568a3e212c01f379aeb3fe0c9ec5a4182bd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 20:12:15.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1012/.","accounts_in_message":[],"_revision_number":12},{"id":"c280fed5e43f28b6351822c7d8a430b419ef38bd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-06 20:32:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/888/.","accounts_in_message":[],"_revision_number":12},{"id":"43d1f74fea303b421dcb009867a04c5f51409035","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-07 11:14:34.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/12/20250607154246/NEC-V-ISCSI : SUCCESS in 4h 09m 33s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/12/20250607154246/NEC-V-FC : SUCCESS in 4h 31m 45s","accounts_in_message":[],"_revision_number":12},{"id":"4e503f3923f7d498e8d83397dab5770329abf1c9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-07 18:58:11.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/12/20250607183701/NEC-ISCSI : SUCCESS in 5h 51m 03s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/12/20250607183701/NEC-FC : SUCCESS in 9h 21m 07s","accounts_in_message":[],"_revision_number":12},{"id":"86cb13816d05d12b9b199b3956b4e05310471351","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-06-09 10:53:50.000000000","message":"Uploaded patch set 13: New patch set was added with same tree, parent tree, and commit message as Patch Set 12.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"49ed63ec78c9e06f5f507e85ac2f6511f3122048","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 12:11:06.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76c0d1888764a0d86ca759c4104b9a2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f7a7705c66224e7bbc65911c39cbd7a3 : SUCCESS in 21m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/40dfe6c986314049b12c4ddc6c836e8f : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d15e5bd933bc4d22870341e4eafe9ea6 : SUCCESS in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"e8840c2058c275c83f51f9d6cab774a5c099da26","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-09 13:04:07.000000000","message":"Patch Set 13:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6f343bf731de4648b46c0b47c6c3539b\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/664abe6f8c9645aaafd503309d0b0ff5 : NODE_FAILURE Node request 200-0000064917 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ed7dc5036cbf4d8c8b1ee85a29415f58 : NODE_FAILURE Node request 200-0000064918 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/58c61f675c03457fa6760485f1540c48 : SUCCESS in 51m 10s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e1b150359efa4638a58d12cf86d6c41b : SUCCESS in 46m 36s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a0f7e9a8c7504f9ebec90588ad57fbd3 : SUCCESS in 42m 04s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd9ec8a8e3fc4567962ebf5fcc7d8ab3 : RETRY_LIMIT in 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fe947f2c5840480e958f1814420808ce : NODE_FAILURE Node request 200-0000064969 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/344974a9f1064bf49ad632c8a8ece02e : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":13},{"id":"2f9af295ae5ca139b763361ecaf043a0821a6f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 13:13:06.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07557676e64740d982a7f11e3b1b071a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ca2cde1cff4468482d8d0425239c558 : SUCCESS in 8m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4d2567f74524f949a4cf2c543573dd5 : SUCCESS in 10m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2911c6271e3444a68dea005e08bde5c1 : SUCCESS in 11m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e1e015ecfdc496faaac030bdecccb2a : SUCCESS in 15m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd7e9560159491dbfb89e24e5a634f9 : SUCCESS in 18m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/d7b9f3d9de6341858f1144e51b4c0a79 : SUCCESS in 58m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/216a3f647a844783b0c64d0192bb2a8e : SUCCESS in 1h 00m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1993eca7fdd442d19ea2799e9c0c2e10 : SUCCESS in 1h 45m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a2341dd668d4a3d9dfccc278eebefc6 : SUCCESS in 1h 03m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41dc6aab1c214010b0817c1f17e39585 : SUCCESS in 17m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d4316eff92c441c1912d40423b924236 : SUCCESS in 26m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0e90d061edb54109b999f00162e66ce0 : SUCCESS in 7m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c3a46a682170417a8cf1d798e2fc7143 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b1f128af6ec24bfeb1936c32b94af7ba : SUCCESS in 9m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f18e67d3e4d14eacabe15f1e837d0863 : SUCCESS in 8m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/acbb05fde6034b40a4bbc5525a8f9bfb : FAILURE in 31m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1384afa7efb74666a4ddf5a73784c19e : SUCCESS in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6f649579edf541b49b19d159eae875dc : SUCCESS in 1h 42m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a4a2d852fe31403d8b591915dcca6a6e : FAILURE in 2h 11m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d7f4f25907f8427c8877cc7157acbd05 : SUCCESS in 1h 30m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/aacfbc5cce4b4383b65a6c622c843632 : FAILURE in 18m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/751588c79460459bb28159aa682936d7 : SUCCESS in 28m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/436d05eadd1b4c93aa259c54aafa3e8c : SUCCESS in 1h 04m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1b5f9feca748499084c62787e9f88bb7 : SUCCESS in 1h 25m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/068c5112d56647cab597c9ee1857eeb5 : SUCCESS in 39m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eb23fced839548b2bb6ac601a848e9f6 : SUCCESS in 1h 21m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/925f8d15b317401595a82b992f795844 : FAILURE in 23m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c631e1f33f254ba2b88e203c79ac46ae : SUCCESS in 1h 36m 34s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da459d4d838f422ca7de52faa1409a9a : SUCCESS in 1h 45m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bf0cc071fcb49fba7d57ab1ad0c9558 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":13},{"id":"0d4d2275eff865d85395cbfbb6fb6f002cf8d0b6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-09 13:49:02.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-13/FC/ : UNSTABLE in 3m 53s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-13/iSCSI/ : FAILURE in 2h 54m 47s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"65c66dcfd94e1a70a87bdde3b7edc40c37e230a5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-09 15:33:06.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"ef97eb7f1e473442148d1840248308a0239fbde1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-09 15:33:08.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"47e23f57dec91a32fa28fa537a90c8b7277d4959","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-09 15:33:08.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"7f42df95666ccb77f10308484748cd8ed2d18b16","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-06-09 15:33:15.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"4b51d66a9ddebf360014f7eb18bf715d76b0eee7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-09 17:17:57.000000000","message":"Patch Set 13:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/913/.","accounts_in_message":[],"_revision_number":13},{"id":"12ce9495b701764affd2d29328574f944905822e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-09 17:20:39.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"6ee48e36e4fde6c822ec00ea19fc6c2aa361d038","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-09 17:39:17.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1037/.","accounts_in_message":[],"_revision_number":13},{"id":"3c4d3ead24a371684da16b4e2e80e5a8bb7784f4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-10 19:59:27.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/13/20250611002443/NEC-V-ISCSI : SUCCESS in 4h 11m 10s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/13/20250611002443/NEC-V-FC : SUCCESS in 4h 34m 41s","accounts_in_message":[],"_revision_number":13},{"id":"a8033841d381e6736f2b5225d47553fa21c3c2d8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-11 09:32:04.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/13/20250611084021/NEC-ISCSI : SUCCESS in 6h 09m 05s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/13/20250611084021/NEC-FC : SUCCESS in 9h 51m 41s","accounts_in_message":[],"_revision_number":13},{"id":"452da63712432377e6391fc353e35d54e9360c03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-12 08:53:15.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"b5b92dfa45a1d3d297de504ca4052a6b6d30cdee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 09:34:15.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f79267e3168e4f6dbf8c6647639ccbc1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c00734851ee4f89ac262de5b98daab9 : SUCCESS in 16m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f447756414064ffa9517cb371bf1c046 : SUCCESS in 16m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6970d36783b244e899cf8d0463d0199a : FAILURE in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"5a9990991b214f288ce5394de07e3b518c33968d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-12 10:33:57.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5fe511c51a2b485ea5e19c3d3d82c265\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/130f962ce2794a528102fd206f6c9873 : NODE_FAILURE Node request 200-0000067106 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/aa7a6e8543544adfa6e0a3b06b4e0a6a : NODE_FAILURE Node request 200-0000067107 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/421dd3a3c5354bb58c6c828d513154ae : SUCCESS in 45m 30s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/16a8b46984b34859985b0742608b626e : NODE_FAILURE Node request 200-0000067109 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6a56154ee6f846bcb0b235f2896152b6 : SUCCESS in 44m 39s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/26cd648d627547bf893113968552edd7 : RETRY_LIMIT in 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/667cf388ac164838822720404f81a1ff : NODE_FAILURE Node request 200-0000067112 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/75dffb394a3b42fbb93565fb323f2189 : RETRY_LIMIT in 35s","accounts_in_message":[],"_revision_number":14},{"id":"e7f4832e6b732f8961840cc440bc365f1cbe7a5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 11:23:20.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed8cb4b2c03146408c896d4ad70b8476\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/738d5035fa3c4f69b8457b5c66fcf745 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2caa121deb1840a88459a44bf06a852e : SUCCESS in 11m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65554d13466f4e2c80aba78e2ebf0e1c : SUCCESS in 11m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5d126ba3075407e8fd96634ce3de8af : SUCCESS in 16m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df64b2700f64fe893aec0e3ec2792c4 : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/b15a971e967840d69cdfbb3e14eaf1ba : SUCCESS in 51m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a9cca1ece3b043039a3ebf29e097c107 : SUCCESS in 55m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5af0ef1acd5047b4b99b6301f6d203dc : SUCCESS in 1h 37m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5220e6c39c2458b8ef4721deadcba35 : SUCCESS in 55m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e31a718f17345e7904dbd2574a14b8f : SUCCESS in 18m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a66a654f6d654993bf96ee4e9a86d848 : SUCCESS in 29m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/20073744fb3f4edf92eb400a078738c0 : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b423c225bdde43ac972ffbdf7b1abd9d : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e0055fae3d2f4e87b2d1b2fa4800efeb : SUCCESS in 6m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e7e0aefd5fe64615987377a21cbc8037 : SUCCESS in 8m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e02e0010363f440bb4d379ab0dfa951f : FAILURE in 34m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e770972365074e9097c22ef80c96f5a7 : SUCCESS in 7m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/37d7f5723a3d4190991ed1747994200e : SUCCESS in 2h 03m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a1c008a0210c4539b72e81ec7dfb0500 : FAILURE in 2h 00m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/03b1651e85f74c49abc50133d9061d1a : SUCCESS in 2h 10m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/de5f367ffcad4570850ad04b45a5bcbe : FAILURE in 19m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ff090bd4805c48e4aaf879015de54829 : SUCCESS in 34m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a31eb3c0b8014cbda873273bf692601e : SUCCESS in 1h 06m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3434c0598268498c804834e0c4ff83ae : FAILURE in 1h 29m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/18f6a122cbff4b40842ce713571b3bfe : SUCCESS in 41m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c6e36ae41db14a18a94847b9a61e24d5 : SUCCESS in 1h 29m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c4908d1d34ec46bc9445e9f1036bc1e0 : FAILURE in 12m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9703c4a2c4c142ffab1598f74bc4cfb3 : SUCCESS in 2h 09m 09s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d912f2c2f45f4220a0d0d5c37f9d4999 : SUCCESS in 1h 41m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcfa36e2a8e54bbfac16c8122ff2d8a3 : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":14},{"id":"a76c8c52c272789fde0f2c6c3544c0f799bf9666","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-12 13:40:05.000000000","message":"Patch Set 14:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/966/.","accounts_in_message":[],"_revision_number":14},{"id":"70ced6fe6b273475a3e3af8aafabbf8d9775087f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-12 13:46:45.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : SUCCESS in 46m 28s","accounts_in_message":[],"_revision_number":14},{"id":"c228c7c5309e5f33da4ee950526605698c8b490d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-12 13:48:11.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1090/.","accounts_in_message":[],"_revision_number":14},{"id":"4f5f40921a40b2ffcfd7947c1bc6348c4612413b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-12 14:48:49.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"415fbf084e88c7131bdb773216168676158fcc7a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-12 16:12:18.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 18m 59s","accounts_in_message":[],"_revision_number":14},{"id":"b085a503405110195e10b636af6f52a02361917e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-12 19:45:36.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/57edf90e738d417e9d9da7f6e736cdfb : SUCCESS in 1h 23m 59s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8d0cdab738a24876806e6514aadbca9d : SUCCESS in 1h 19m 53s","accounts_in_message":[],"_revision_number":14},{"id":"a47ec284544b261a75a8026e5ad19bcb54794552","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-17 12:28:44.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/14/20250617170110/NEC-V-FC : SUCCESS in 4h 23m 05s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/14/20250617170110/NEC-V-ISCSI : SUCCESS in 4h 27m 31s","accounts_in_message":[],"_revision_number":14},{"id":"05976aa35ad43c0404f658bd4c6669d6b71bfaf7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-18 03:24:43.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/14/20250618032002/NEC-FC : SUCCESS in 6h 17m 16s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/14/20250618032002/NEC-ISCSI : SUCCESS in 9h 04m 39s","accounts_in_message":[],"_revision_number":14},{"id":"b727c9df43cf8503b6af84e9fc0227b7397d6714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-07-08 08:33:46.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"5e72dd9c95e223fcb6eb4b229733147ec694e43e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-08 08:34:55.000000000","message":"Patch Set 15:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":15},{"id":"151da2f4f234ff71fc09fc6adc91518c0bb154cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 09:06:14.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd44d24af7ba4120b76865745c5cc659\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18fa157c0719455b8589154e1c07213e : SUCCESS in 19m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7d7750d42614a6dbfcc8d2a1aa78a92 : SUCCESS in 17m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/58c87c64f400414b8fe06c99b05bc07e : SUCCESS in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ebef734544a3b5eb31ac142287834e6243bb19c6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-08 10:19:44.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : SUCCESS in 45m 50s","accounts_in_message":[],"_revision_number":15},{"id":"6d60cfb61b01c813cfa1a56442883c7c969b5329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 10:34:06.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/119eb21a3e24440184fefcca7f5311ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f29bf1ab64f4281a349354887ca2311 : SUCCESS in 7m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bce609c539eb483aa0ce9fffb6c95307 : SUCCESS in 11m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ab072714a84495bb1c0b8eab7e1ca45 : SUCCESS in 11m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bbf9e58ec5f04fbba0eac975154779a4 : SUCCESS in 15m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bcf95681f8b4a169030904da562aea5 : SUCCESS in 15m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/97b1068576c24b65aa89a0d4e40cbc60 : SUCCESS in 1h 00m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/378df1bba44e4006942534e4d74979c2 : FAILURE in 58m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3fd4356076474451811d9d7e50073a3f : SUCCESS in 1h 40m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c941bdee5566404ca2431a4307bade26 : SUCCESS in 1h 03m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58d0d972984349899d8909e0208fe516 : SUCCESS in 18m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5cd9b544532a40189535360bb0488278 : SUCCESS in 24m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a131363f5dd4488a99d226e659b9bb83 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6f3ce636dc7d40df83ea8eeeee57f92e : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ed4a1c87c7e047aa8fd0ba3c2dba8043 : SUCCESS in 6m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5a6c3d77a3254a1881e47513f2a4acbf : SUCCESS in 8m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ea71d6e620af46a8b0cf3f3dd5d1917a : FAILURE in 36m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4913892fc05142fab4d915fb98c4e23b : SUCCESS in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/72b20b562c1141918249910e62fe9b7c : SUCCESS in 1h 55m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e4982db6180440a29a2104f523adfbca : FAILURE in 1h 36m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/950a6f0a8c6b410e8c74af7d3f00f4d6 : SUCCESS in 1h 48m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bc7f6c56b5ad41b9adec0bc1574892c1 : FAILURE in 23m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/44f8485af2f8411ca444f8d5a238c2b4 : SUCCESS in 32m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/81bb19cc4f7445e986d0732c2fbb0fbc : SUCCESS in 1h 10m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dd5b432b9cbf4533bde3d9f48a71f7ef : SUCCESS in 1h 19m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9ebfe88e57304628beef37c2141ac247 : SUCCESS in 54m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/20690b15b7be4da99f4200b277031ff9 : SUCCESS in 1h 15m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/eb172cfb60d141d29dc8674244434929 : FAILURE in 12m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a662f2ef98a74bc1a2ea6ecbcc0757a3 : SUCCESS in 1h 40m 12s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76380cc5d4fc444787d80ffa0ca19752 : SUCCESS in 1h 25m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c757085ba51844c38ff31b651767175d : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":15},{"id":"c779a9af1c805148cd06872f194fc188835f2277","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-08 10:37:07.000000000","message":"Patch Set 15:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1184/.","accounts_in_message":[],"_revision_number":15},{"id":"25f6b4381273e5f390e039bec0a29e65d663c012","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-08 10:52:45.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":15},{"id":"e35c3abf447fd9dd4f029980d3ceac2095b699cb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-08 11:03:24.000000000","message":"Patch Set 15:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4d2958bc5f4a4dbb86b90341bf10b2b7\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/61c648c424c24b828c7710ce8b8ffa18 : NODE_FAILURE Node request 200-0000071929 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ba4be7f48d1b4bdc894848451cf603aa : NODE_FAILURE Node request 200-0000071930 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2d56a3061c0149cfb145440043156833 : RETRY_LIMIT Host unreachable in 50m 15s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d265b566954c4c7ab413647457f6ed62 : RETRY_LIMIT in 30s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2aec8398f0ab481bb09f596a904cc3f6 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":15},{"id":"b3352f6b569bc4ac73edc02f743c13f905b12713","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-08 11:35:31.000000000","message":"Patch Set 15:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1314/.","accounts_in_message":[],"_revision_number":15},{"id":"38e1550025ef6575f6a15bb3ea783066d37303e3","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-07-08 16:01:44.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":15},{"id":"e2dbd919804cc792022a79ac8d0718ef8794dd97","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-08 19:50:47.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"28bfbc3da14d91c8d53578b466c55d0fc6402797","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-13 05:41:30.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/15/20250713053358/NEC-FC : SUCCESS in 6h 18m 35s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/15/20250713053358/NEC-ISCSI : SUCCESS in 9h 07m 28s","accounts_in_message":[],"_revision_number":15},{"id":"fde8bf2c3fb760d9737f4a39a0cc1c1fa6660f79","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-14 09:34:33.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/15/20250714125426/NEC-V-ISCSI : SUCCESS in 5h 19m 59s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/15/20250714125426/NEC-V-FC : SUCCESS in 5h 40m 03s","accounts_in_message":[],"_revision_number":15},{"id":"8a3c81b7f0a4bdcb4cdb5e1eb55c0f1bde04128a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"real_author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-07-17 10:07:33.000000000","message":"Patch Set 16: Patch Set 15 was rebased on behalf of \u003cGERRIT_ACCOUNT_36800\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"}],"_revision_number":16},{"id":"d63fdad990023890ea90e82bb440ecd15f0daacd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-17 10:08:51.000000000","message":"Patch Set 16:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3e5c132c2a6c4218a5a9b2f45854ce94\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1b6a2a7446b04f3fbd5c0a78a9981cb4 : NODE_FAILURE Node request 200-0000074046 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2e0ba58633be4b00ba7a7e471ce7dda1 : NODE_FAILURE Node request 200-0000074047 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ecd8cfafb7a49219c73b43521a0b2e1 : NODE_FAILURE Node request 200-0000074048 failed in 0s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/72d65fffc92e4d1a9d50d9940f7d4bcf : NODE_FAILURE Node request 200-0000074049 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/37db2de56e53448e81695ade510c6037 : NODE_FAILURE Node request 200-0000074050 failed in 0s","accounts_in_message":[],"_revision_number":16},{"id":"deca03183e37725429babc16bb7ef1e09b07626b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 10:43:31.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db8032225c864233ae433a2a6b780edb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e90ac726427746298594e3c0446e5a5e : SUCCESS in 26m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1a1df2254e6e44eba99e37fe84041464 : FAILURE in 29m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/48299a771e574b82874af3dcf04d4f90 : SUCCESS in 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"229c67ea96d09794b0cffd835c3347a4bbce4fcc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-17 12:05:53.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b53e33b72d134b7ea4905fbf1e4644da : SUCCESS in 53m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/658a50b99f35495c93bd458456d5c5a8 : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":16},{"id":"8232f34ef5fefc89d07af98ff1fa128bc47b0b29","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-17 12:19:13.000000000","message":"Patch Set 16:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1241/.","accounts_in_message":[],"_revision_number":16},{"id":"3e6241b9ed7101c7aa7fddd9ed930dc2b3b2b28b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-17 12:19:55.000000000","message":"Patch Set 16:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1373/.","accounts_in_message":[],"_revision_number":16},{"id":"bf7e8679573524bc9a4821e21975e381a3c31369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 13:13:14.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0349a952260946329e0c3eec5e42d21c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ace8241bf0a47b48a27ada2aca6a571 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8392f75a2c54360a8bf7a619129973a : SUCCESS in 11m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a65f01cd6a8048d194c047011963a8e3 : SUCCESS in 7m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/647093c52755416f8447edac6d090d1d : SUCCESS in 13m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0194a95e4c6c48eb90e3e2bdd0c31018 : SUCCESS in 16m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/d7e373d7bd44436c9207fe718f5bd20f : SUCCESS in 1h 09m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9a8fbdaf1eb1447b90ffaf553ab961fa : TIMED_OUT in 3h 02m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/744d703125e04c2899ffcdbe1a359e4d : SUCCESS in 1h 50m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f296dc4c59464e00927510859f1d126a : SUCCESS in 59m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb71ad3e54f947f880e4b196bd7c9a4b : SUCCESS in 15m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/056c76e4bd2f40e284f853971ae4879a : TIMED_OUT in 41m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d2394344b25946ad80c4de5e256517e4 : SUCCESS in 7m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/726ad6565e524e1baeffe8657110d54e : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e68fa55b2f5e4c818ac7c498ca85adee : SUCCESS in 6m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fdac8ec49f1145d1b2f9827357ae44c3 : SUCCESS in 10m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/08e5af3734af459f8f23b9d8a054de44 : FAILURE in 42m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6163c62b8cdb47b89a944d3c4441ecd6 : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1e5421f27ffa43abae4f150baa7d1b1c : SUCCESS in 1h 59m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bab50f15801345e9a5311c0e9e1fd113 : SUCCESS in 2h 27m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1140cef7d4e442f89274aef384ffc9ee : SUCCESS in 1h 36m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5d20a38c08034c2c9b3dbf3594ea79b9 : FAILURE in 22m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0358757719e944bfbf56b6ff25242dda : SUCCESS in 36m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4cdd934137aa42daafb057c1500fd095 : SUCCESS in 1h 08m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7bdd067052dc45618815743ac7aae8ec : SUCCESS in 1h 35m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/11d9fc0b177e44cf9d2b7af0c5e0aca4 : SUCCESS in 42m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/47cdd4ec70324aeba13b2ea4d7f5c470 : SUCCESS in 1h 30m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/51dc069fd9744f9e90d127fc1473a664 : FAILURE in 18m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f011effdfe814696876e6d2ff7d42ab7 : SUCCESS in 1h 44m 40s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9f325740cac45e8af7e89835a4b41cf : SUCCESS in 1h 34m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7ea0ce459374b62b111981ced97e3bf : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":16},{"id":"f761197f0374755a86d6bcf43efb82ad3dc6fb62","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-17 13:45:56.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"baa2fcad4f4c7dc721be8da16e9d1f92280c8975","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-17 14:33:32.000000000","message":"Patch Set 16:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-16/FC/ : SUCCESS in 4h 25m 40s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-16/iSCSI/ : SUCCESS in 1h 49m 21s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":16},{"id":"b5821511a557bad80f8af3fc771047301dd6d9c8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-18 10:10:14.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"678f3ff53d967d83440d5cebf2d7231e3045a8dd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-18 10:10:20.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"34f07bf0c61b627bd69037cba362424bf02c1f60","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-21 08:06:07.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/16/20250721080849/NEC-FC : SUCCESS in 6h 18m 18s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/16/20250721080849/NEC-ISCSI : SUCCESS in 8h 57m 16s","accounts_in_message":[],"_revision_number":16},{"id":"2b63d0e6b9bec87f2597c6bb25ea762b0e0d101d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-21 13:14:03.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/16/20250721161007/NEC-V-ISCSI : SUCCESS in 5h 25m 13s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/16/20250721161007/NEC-V-FC : SUCCESS in 6h 03m 52s","accounts_in_message":[],"_revision_number":16},{"id":"d5b8f1ff72df09763404747cdc05d71a6e0e1551","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-07-28 10:13:20.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b31f7e0ee27ad5544e346b2f59407497f0474eb9","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-07-28 11:19:06.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"42e3e4c3d359c0f6032ba30839f6ed70c911eec2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-28 11:26:10.000000000","message":"Patch Set 16:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/48/","accounts_in_message":[],"_revision_number":16},{"id":"4137204c7e56a88df85788118c10a613d9268c23","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-28 11:36:04.000000000","message":"Patch Set 16:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/16/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/16/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":16},{"id":"c947828ae2aed82005443345483b0d39e7dbe7a3","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-08-14 14:55:57.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b21306e83c2706b693df71bbb68b16b9ee6a3f66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-08-21 03:32:09.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"83996f54a465b8a044984eed1886fdc94c8da67e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-21 03:35:41.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":17},{"id":"9858386844262a93b109dd4ff9c1e96da3a4074b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 04:43:19.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f43a40633d284c4daa5afe745348a12d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/613005c942d14abcbc510160937180c4 : SUCCESS in 24m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29dce88456e54cea9077cc84d65db671 : SUCCESS in 26m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d385416a102e4110ae92bcf76e6934a3 : SUCCESS in 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"94f465d3642677f25bf39342bb6d19fee12ec6c2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-21 05:03:53.000000000","message":"Patch Set 17:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/239/","accounts_in_message":[],"_revision_number":17},{"id":"357ea42a2654c0c445bcfc57b635fdec6227e9ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-21 05:19:45.000000000","message":"Patch Set 17:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/17/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/17/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":17},{"id":"4a4f5631f9d9a0227a3881d73324b85264c77533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 06:13:28.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/792165b3840d4c828c5e98d8cc466062\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c962b0a16340959acdd25bdc203874 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93e712d7755e4528849b7212dd21a98e : SUCCESS in 10m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5abc17f2df52466ea761dcd87c1512ab : SUCCESS in 11m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/013ff6c9fab74623927dd7fea70a4944 : SUCCESS in 15m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1005c601e4545c1be24ddd9dac0e571 : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/676d07fe3b7e4729b590b484d347bc31 : SUCCESS in 55m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ffef5f85f43f4641b7eedfcae1fdcbd2 : SUCCESS in 55m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e327e014fbdc4eb8a72757a7e11fa8ff : SUCCESS in 1h 25m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49e80273d8674ceab7beba4611ae7f61 : SUCCESS in 39m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f31481a585ff41dc9af991fb40da65a4 : SUCCESS in 16m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e0b5a92effcd45aaa7feaa6a1893c17b : TIMED_OUT in 41m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ca910abc663b40cf873d883628fc7541 : SUCCESS in 4m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6d02c2676f4f43c395f663d0b68f5fbd : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3e99d1ee0af145eeb0af880d373b598e : SUCCESS in 4m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/823dc0202239427dbca5156e97a1cd04 : SUCCESS in 8m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/37e808b8ba5b438ea39c345886b85d53 : FAILURE in 37m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9b5a85d35ff144d8a896032b6eeecba8 : SUCCESS in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6c63e3c6ef2e4d899a735d084751987f : SUCCESS in 1h 41m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8cf95112bfef46a083e2da8f50cd48c1 : FAILURE in 2h 35m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8261d097f559470caf18394c6351e790 : SUCCESS in 1h 43m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/af775dcdaa134ba592eee1f7340f4e3f : FAILURE in 18m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/05279547d81c45a3a91062bc5e063f63 : SUCCESS in 21m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cb29dbe141c34176beae64dd66a11ae8 : SUCCESS in 1h 07m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e333bfc05e6641e6a79a6a1524dccaac : SUCCESS in 1h 13m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dfcc5f120c044919b0a50c97459a7eab : SUCCESS in 38m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e0d4b30d08fe4d14b648662eff9fd25f : SUCCESS in 47m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dbbeac6d42f947df82bf8a44c4b3a860 : FAILURE in 23m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8cf8e40d153d4c5398293ddfe0eae143 : SUCCESS in 1h 53m 03s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16f3c9672a3c4b5e8181a90f7cc7d871 : SUCCESS in 1h 25m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a8b3bf5c26646a99f9f92c35582eb08 : SUCCESS in 34m 56s","accounts_in_message":[],"_revision_number":17},{"id":"06e358c344f5e6a0b4f5bf2e6bd30aaa46522484","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-21 10:06:35.000000000","message":"Patch Set 17:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1510/.","accounts_in_message":[],"_revision_number":17},{"id":"19ba416f5f78a7760910e09c270e99cb84bd829f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-21 10:44:40.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/93a3e0a69e724477b18eba3e7a295fe7 : SUCCESS in 48m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fdd27d23403145ea98d8dbbaed95c481 : FAILURE in 1h 14m 43s","accounts_in_message":[],"_revision_number":17},{"id":"57799f3627a4b5b30bdb2b9fa04733f642231277","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-21 11:22:37.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 25m 24s","accounts_in_message":[],"_revision_number":17},{"id":"f88a50383ddd841168644b1e90197e00a0b47830","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-21 11:54:46.000000000","message":"Patch Set 17:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1645/.","accounts_in_message":[],"_revision_number":17},{"id":"880df5559ab57ca54eec07c69454aa22fc72ff8a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-21 15:25:37.000000000","message":"Patch Set 17:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-17/FC/ : UNSTABLE in 7h 54m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-17/iSCSI/ : FAILURE in 11h 50m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":17},{"id":"a0d7d9049a5f29251e32d70878e1b9c940d2b2e2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-08-29 05:59:07.000000000","message":"Patch Set 17:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/17/20250829045657/NEC-ISCSI : FAILURE in 10h 00m 48s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/17/20250829045657/NEC-FC : FAILURE in 10h 02m 07s","accounts_in_message":[],"_revision_number":17},{"id":"23d6085e62f0fa6ac38663fcd2d2cd4f2ab57dca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-09-10 08:34:13.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"24f5e8957d4157174ee6bc8d47751e4e136a6ea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-09-10 08:37:51.000000000","message":"Uploaded patch set 19: New patch set was added with same tree, parent tree, and commit message as Patch Set 18.","accounts_in_message":[],"_revision_number":19},{"id":"d453f268375975ff87d9190485884460f4d68df6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 10:08:46.000000000","message":"Patch Set 18:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/526/","accounts_in_message":[],"_revision_number":18},{"id":"948219597da3ab7c482baa521f06bace854581af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 10:27:46.000000000","message":"Patch Set 18:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/18/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/18/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":18},{"id":"fa3c8fdfcf42c92dde561c89fb80853e444dcd45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:28:58.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54cfbc59dc7341f0a63199ae1ede4f7a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ecdc44a4da9d43bf96ab8e6fd81d2264 : SUCCESS in 21m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c2de6716e78e4240ae06deae2576766d : SUCCESS in 24m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/638329af9c0449e1b1ca9c153bfbc791 : SUCCESS in 30m 57s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"94011cf2dca288f35732ca91941cf5d9b9d479b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 10:46:42.000000000","message":"Patch Set 19:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/528/","accounts_in_message":[],"_revision_number":19},{"id":"f23cdff849a1d0a9ac83e1f914cd55c7bd9667c0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 11:06:12.000000000","message":"Patch Set 19:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/19/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/19/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":19},{"id":"0eacf7e6174fa61b2642bc5bf2d795a455c64879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 11:31:08.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64c2f54a550d4b85a0bab0e6a8b88af1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f5b500ee8b44c9a8ddbf92e05fc6db2 : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/424a662477db40a3ae1e30e1ecab5885 : FAILURE in 12m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c1a49a22f794297a5b9061fa653395d : SUCCESS in 10m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5eb5f20cef654905bb6d8edaf5523482 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9206977494564560a6ee8865b9c3bd49 : SUCCESS in 14m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/8491cf4a3b374a35b0cdc508d1c6b2ae : SUCCESS in 36m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99bdeacff0184999b4a3c589c5e43e15 : SUCCESS in 1h 00m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dad1a8d9e95d4bb88662cb7fcca9eb96 : SUCCESS in 1h 14m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8dfa5e2d2824eac9a0f671e1b14e258 : SUCCESS in 1h 07m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/222ec854521644b7b5cc5a1613431794 : SUCCESS in 15m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4f2caf240b894634b3996f4f988189ad : SUCCESS in 21m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bf978e5f9ef44198a62ec2dc73a7b73b : SUCCESS in 6m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5840fee0a3224a59a918d7d866210e62 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/592b916d306b47bba215bb39eb5f29d6 : SUCCESS in 4m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c6c4d25397444b60a861233d3b8e4c39 : SUCCESS in 8m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/17c88ab0af73463e93c2c507945ecb7d : FAILURE in 18m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/87dabbdb955a48ed9c4422295acc0b57 : SUCCESS in 8m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a7f10e510374270b9c9cca07ec74e87 : SUCCESS in 1h 50m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6c2758293f0e4d8a967beb972acf395e : FAILURE in 2h 37m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/672e5b9c31334d3b970cc9c8776a8a8f : SUCCESS in 1h 26m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/60570dfc320a4071b96e60b2507f8499 : FAILURE in 20m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/203e72f2ff884d3694e820dbc1e35bb0 : SUCCESS in 17m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8317d8b4da3e438390f38aabb5882888 : SUCCESS in 1h 07m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d796fecea2bd4d83b292cece823c3ba5 : SUCCESS in 1h 35m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ed0893cd5cfb4424b15366067196f025 : SUCCESS in 28m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/962ea3b6089e4e6a809d17a0348c5d04 : SUCCESS in 1h 36m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f8e7b1bd71b74b8ea613c5c857b26d32 : FAILURE in 24m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e04de9410c3948de9804559312bed08d : SUCCESS in 52m 34s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5016cce8f74a462ebc9202a7b8b524dc : SUCCESS in 1h 12m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37d34ad3bec047fc9a9796e2b04d3619 : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":19},{"id":"a4055a9cfa20a12679fb48c601033be2303ab4c6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-10 13:46:28.000000000","message":"Patch Set 19:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1744/.","accounts_in_message":[],"_revision_number":19},{"id":"6b5c0c90a469856e3847391b8bd7b2fab7a67d6c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-10 14:27:57.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9c4d336b1b684dadbb1661aa9a756248 : SUCCESS in 1h 05m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e5be53328bc846d88d20627ec7e45300 : FAILURE in 1h 07m 02s","accounts_in_message":[],"_revision_number":19},{"id":"51251a7f48bdfd2ec58a5de2a59892987b4c94b8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-10 14:30:32.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":19},{"id":"3da65b47aaeac1765fc904fc668f5f94ef91914f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-10 14:59:03.000000000","message":"Patch Set 19:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1878/.","accounts_in_message":[],"_revision_number":19},{"id":"7cce262908012de272b5878bc53efa551530e1ec","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-10 15:27:23.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 21m 41s","accounts_in_message":[],"_revision_number":19},{"id":"115ec0b25463a2c264a256eaf5b8503954d1d44c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-10 16:10:24.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,nlogs at https://oscilogs.quobyte.com/logs/$JOB_ID .","accounts_in_message":[],"_revision_number":18},{"id":"fb3f09160e91bc14fdc2bcfebd88cd96a416d4c7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-10 19:50:28.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,nlogs at https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":19},{"id":"defe8311c62a9280594861b405fa49549dab4786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-09-29 06:04:21.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"e256ba025cdbbbe1134d3f96c07f5dbb4f5482cb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-29 06:05:05.000000000","message":"Patch Set 20:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 938771,73","accounts_in_message":[],"_revision_number":20},{"id":"765ac9576cee1bb89a33d1749682d968e1a241c3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-29 07:21:42.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 36m 12s","accounts_in_message":[],"_revision_number":20},{"id":"10d31e7ab2cb7546f72a802b2304328b21d060cf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-29 07:23:30.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949483 : FAILURE in 35m 40s","accounts_in_message":[],"_revision_number":20},{"id":"a52b7078f0b5c53a2d744b6617e4dbc79e11f5e0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-29 07:56:27.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":20},{"id":"00e6811989f062e1750b380d3292846f016ea78c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-09-29 08:33:54.000000000","message":"Patch Set 20:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-20/FC/ : SUCCESS in 2h 29m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-20/iSCSI/ : SUCCESS in 1h 41m 35s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":20},{"id":"0941796aef422706765b4b5f310a171d4ba769e2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-10-05 20:54:27.000000000","message":"Patch Set 20:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/20/20251005192806/NEC-ISCSI : FAILURE in 10h 04m 34s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/20/20251005192806/NEC-FC : FAILURE in 10h 26m 18s","accounts_in_message":[],"_revision_number":20},{"id":"e45ecbee63b49685b05f4ccf203e15d432860502","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-10-09 19:06:48.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"4c3bb7ef1d00952edb3071ea04ca7b6194c20928","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 21:17:10.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e30a268c32f5448291e142c29f68d6d7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/72ced04c52754068848f9e4b6c0e2b0a : SUCCESS in 24m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa099350d81d49edbdc74fe78926b7e8 : SUCCESS in 18m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9899ddabd52744428ca67e13ae4813bf : SUCCESS in 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"a56e5512d20805e1e67516ee2c5b602fc7a7a36a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 21:38:48.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1356a61d09db4d0b80ccd619b41cb70a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d51374d7384e4759b53cb0ad8f8777ac : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b755ab03280542b39e8f3076ebaa2d6d : SUCCESS in 11m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6e638470bcd4541b4a2c64bdefa3c88 : SUCCESS in 7m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c69a13b785e49d8bdd034a692d62437 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13321af9dbe64dabb3ad88694c01430f : SUCCESS in 16m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/003eb314c15c416fb9ad56522529f0d1 : SUCCESS in 46m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d2357a3602864c42aa07ec3062a4397e : SUCCESS in 54m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2392b05aab6c4736a23805d0decae8fd : SUCCESS in 1h 32m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee2f7c88ce7c428fa4f30e6d60739ce8 : SUCCESS in 56m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4d0219615b24377b6c7c8751b8edf60 : SUCCESS in 15m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1f145af16b9e49108a0c040a7b38315b : SUCCESS in 17m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/047e4804574049ef880ba3fc1da8c55a : SUCCESS in 6m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff77d63505c44a46b4471d5b3263cc79 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f07ce053ea6843b0a50a618783a78ffb : SUCCESS in 8m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9c90faa589f34b878b0981a7a286f52d : SUCCESS in 8m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d466f136382e4a59825e1ab45f872b01 : FAILURE in 19m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8bee30fc1ff14adfb6e0e40c6ae8c24b : SUCCESS in 7m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0ec304c1e2754603b867e8b9f1e7c9c5 : SUCCESS in 1h 45m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/632068712079477eab16b0c7330ddbe7 : FAILURE in 2h 25m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ecb90b9588764f6b8fb64b6540073cd8 : SUCCESS in 2h 10m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2852c1a0921645dd9a6394356b440a22 : SUCCESS in 2h 04m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fa011122aadd430887a04e9ee6439b13 : SUCCESS in 35m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/77be3f91b0d5471fbc184325a8efbc85 : SUCCESS in 1h 08m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b06c3a30ef654591b32bf8f33d211f18 : SUCCESS in 1h 51m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e38f881075614b369883511885860dee : SUCCESS in 44m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7118d7aa319e4ad9b151a60598234e59 : SUCCESS in 47m 55s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ccec0600ac1b451d8d3267c21418afb7 : SUCCESS in 1h 14m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40101d07cad8440bbd9987f0ae8b26bf : SUCCESS in 1h 33m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ac2e2390c1f4094bb7afea1641edd91 : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":21},{"id":"8b99d23a0f50b9b8cba013bf4ea0755c3894305e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-09 21:43:56.000000000","message":"Patch Set 21:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":21},{"id":"3fd867abd3c0749097f1fb5b403eca032b3780fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-09 21:52:41.000000000","message":"Patch Set 21:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_10_2025_21_46_22\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":21},{"id":"048d5b84f88f168a16b85534c398ff10c2b9f290","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-10 08:13:37.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":21},{"id":"b3e636bfd2f7d86a58d2138f9d9e91c922ba3ce3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-10 09:57:50.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0d955caab1ca4b5eb9f30f2f550b5116 : SUCCESS in 44m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8aaf06f0b74b4c23a9ee6f0eb70e13da : FAILURE in 50m 23s","accounts_in_message":[],"_revision_number":21},{"id":"02e9af281a6ca74dfcaea270ed1b461e56e51d3c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-10 13:53:52.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 13m 07s","accounts_in_message":[],"_revision_number":21},{"id":"814f7b38ada9a91d1efdd0e751b8ca04744ddec0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-11 11:05:08.000000000","message":"Patch Set 21:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1992/.","accounts_in_message":[],"_revision_number":21},{"id":"6d39488019e9e753c82760b667b84274ddf7859b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-11 11:36:27.000000000","message":"Patch Set 21:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2106/.","accounts_in_message":[],"_revision_number":21},{"id":"ff095cd7b1a86f95417953962ec11dcd59aeceb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-10-16 07:08:57.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"62a8489ac9e4b8844b499d4f43da6cb1364169d6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-16 07:13:21.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":22},{"id":"47d30476aee05d2c92e5702d08f71df11148563c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-16 07:17:59.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5aa01e68ceb7412596a94fb32c5445cd : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6c74e71dfae941428bb8d05aef65ee00 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":22},{"id":"82dd0d8c78ad678d53511693731a1417164c34e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 07:57:37.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50952a8fbdba4f5e9c2ca9eff49cd317\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01766cf159244a3b9b9241c52ac0cffa : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c3da6b5140e345eb8595ab13e5b4fd08 : SUCCESS in 16m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0078338deaec456e8b8913bb1ac60217 : SUCCESS in 36m 24s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"d3196549b9a28b3956d61714749f013c6e417cfc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-16 08:09:25.000000000","message":"Patch Set 22:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":22},{"id":"37bd9437d83adcaf54c8e187ab14b15a70178569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 09:52:51.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05df335e9c5a4d43986616b2b9744324\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae297502326d412db2f6e5bcd3657754 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9925698c82bc493d8bdfa62988178dcb : SUCCESS in 11m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/229e4fab09ab49f8b08e9118c5c675d1 : SUCCESS in 14m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39b93c8bd56f4b579e75c2bcd38d22d1 : SUCCESS in 15m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/36250938703d4740b0fab1037bc456ce : SUCCESS in 1h 10m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8ceec9ead61f4d94a3032a315e30f5ea : SUCCESS in 1h 03m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91c9befea5844d358f35ffff141428f9 : SUCCESS in 1h 36m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb307c8161a049b6849f60cde71c3182 : SUCCESS in 59m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38b47d4c682d41aca3f9ea87f2223943 : SUCCESS in 12m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8abee42c0c204a948da440d4e60f43d3 : SUCCESS in 24m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/967e30c8498d4fd4a83d4d05ac4618df : SUCCESS in 6m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b2cdec9ed1334ec4a1e5afae9b60e4bf : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7197b9247fb04b1992f46412a34335b9 : SUCCESS in 5m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a21c7620ecdf4078ba0732b3137bfed9 : SUCCESS in 10m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/93852e2a32b147cbaa54fd74122c0371 : FAILURE in 30m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cf42498075704f43b469c644190c7f68 : SUCCESS in 7m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c32540f855e4414ab8a493b9836d0d78 : SUCCESS in 1h 55m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/143c987543734c24ad024d3ac42d9967 : FAILURE in 2h 29m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/937d045d01734b7bb4edc7db090da67e : SUCCESS in 1h 42m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c01f877ce10f4d49915127a2ac2a7d6d : SUCCESS in 1h 51m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b453477f242f43a7badb9a1d1e19e693 : SUCCESS in 36m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/037918011345461ab2eb3b25d727b171 : SUCCESS in 1h 13m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/49db2fa0bdc84777b86d580d68aeb39b : SUCCESS in 1h 45m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b5b4b877b58d40e5ad8fb13a0edae931 : SUCCESS in 49m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c68d04e94c5746beaa41ab6b46985dbd : SUCCESS in 1h 30m 56s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/b5604de250234237b5707813ce608743 : POST_FAILURE in 2h 13m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d986ce9ef6bb4706b7c699eb01a8f231 : SUCCESS in 1h 35m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54f4e5f221784d6cb07837e842e6a1ad : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":22},{"id":"f3bf1fb368082762dc28649417de101678f3423e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-16 09:59:26.000000000","message":"Patch Set 22:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_10_2025_09_53_02\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":22},{"id":"ba26d8adc5b95465b82450f9fdedeb3b0c3f951f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-16 11:04:24.000000000","message":"Patch Set 22:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2161/.","accounts_in_message":[],"_revision_number":22},{"id":"815a29c944f2e1f9f80dba9d4d04d1a1d5049750","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-16 13:30:45.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 12m 22s","accounts_in_message":[],"_revision_number":22},{"id":"250c079a95da10ba5694b043988493fb45ee85dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-10-17 07:18:43.000000000","message":"Uploaded patch set 23: New patch set was added with same tree, parent tree, and commit message as Patch Set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"808dc8d6b7ed76594a52f7e41625b690d966d787","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-17 07:23:08.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":23},{"id":"35a813974ef61c91e5afd0def304aa4c7e6a1b86","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-17 07:28:13.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c83922e9967f466d9b22944140d8704b : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2186da2d172046b09bf87c2be43afe12 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":23},{"id":"1962451098f561af5825daabc31b01fe34612d52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 07:59:09.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b042662cd9c47c780fb09aa0217d80b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/570cecb4453b47bf98f51631cf008c69 : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e6674486afa641dda32f531341fe5e93 : SUCCESS in 24m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/17ed80e0f75c4b45bd97da473a7700fa : SUCCESS in 32m 56s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"0b54d09439893fe82b49e74fe1f0154d8829bf7c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-17 08:19:24.000000000","message":"Patch Set 23:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":23},{"id":"7c0c96e0ff0d1b590edc8964234fc1c771edcc11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:32:10.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37e0b106cec547a7841dae9058851ddd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c255a19b41042f8b07c518420944436 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef458ba0ef2343869fe2e0f047313291 : SUCCESS in 11m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2bbb413004041bba3116e159d70f07c : SUCCESS in 14m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64a11d8520a49c19a5e24543c0e3c31 : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/a9f3f408c15a49c19d8e48b1ff116990 : SUCCESS in 1h 04m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/402668edd59446258ed33a37f7db68bb : SUCCESS in 56m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5733fd92eeb44fa9a98519acf3f0cce : SUCCESS in 1h 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44f387c790bd4e1994fd6851abf1d8cd : SUCCESS in 59m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5bc1fb8e72344768ec444b9504d1494 : SUCCESS in 17m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6109f0a631ca450d896940d64d3a5f8f : SUCCESS in 20m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/13dcff712138488bb3998e03e5339cb8 : SUCCESS in 5m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b493a1fdbe0b423ba2a0e735d0a7a886 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/91a2a1d1e03449fe851d90d70d3d3d3e : SUCCESS in 6m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3529a357975a4d27817f906f7e8c5b55 : SUCCESS in 10m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b7402636a2344c639449c80220d2588a : FAILURE in 32m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a664f9da46db4b479be0d86b9a21af73 : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3daf4d47a7ef4004a6a253133736691e : SUCCESS in 2h 05m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e4ae871765be431f8fadde109d3e8a6c : TIMED_OUT in 3h 04m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b6ce7a244e7e452385aa6324f3e882df : SUCCESS in 2h 11m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d68f510678634b628c41e6dcaaedc103 : SUCCESS in 2h 12m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0c2cc3e7f8cd4edb9a12a46a8b1e674f : SUCCESS in 39m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/622b6ae91dab4dbb99c9f8028eb486a8 : SUCCESS in 1h 11m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/46a590820e524e1483e15575bd4d44db : SUCCESS in 1h 40m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ef816adfed5e425d891872a805e1e53c : SUCCESS in 49m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c98bd3b428e54823b4f39a3872b9e3e6 : SUCCESS in 1h 47m 44s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/447d92675a204ac0837d0b83f107d991 : SUCCESS in 1h 51m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/27d69d1127ef4afa8a90d48adae327e1 : SUCCESS in 1h 50m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8c21ea6c80f4aa699ec34b35289ebef : SUCCESS in 47m 26s","accounts_in_message":[],"_revision_number":23},{"id":"fcb67d928cd85fb68b24bdedf9abb804398527bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-17 10:40:47.000000000","message":"Patch Set 23:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_10_2025_10_34_22\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":23},{"id":"bd39b1fe34a55b010921669067bd0d8faf862f27","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-17 13:26:26.000000000","message":"Patch Set 23:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2179/.","accounts_in_message":[],"_revision_number":23},{"id":"e6441d96b10bbc0efec2d84c9378f717f6b9a121","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-17 13:39:10.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 13m 28s","accounts_in_message":[],"_revision_number":23},{"id":"1d04076becefa79923cdbc1122e359f85437a517","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-17 19:10:45.000000000","message":"Patch Set 22:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2047/.","accounts_in_message":[],"_revision_number":22},{"id":"9f47bea070a204f309b1dd4c9fb3ede4a7c8612b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-18 12:35:11.000000000","message":"Patch Set 23:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2065/.","accounts_in_message":[],"_revision_number":23},{"id":"1cc00c52529217d6d2a61aed9e5df91a3a626851","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-10-31 06:57:42.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"3f3e673bec822689fe449b939898c92f33e18bc8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-31 07:03:19.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":24},{"id":"cfdf1749639f4321c7ec97644cf87e6dcfed9dd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-10-31 07:07:32.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"55ce22478a0eb3f8f04fe79dda8cc35c5a2a990c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-31 07:16:14.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":25},{"id":"87eba41bd3a62262167df2a9a84b0cc570c53e51","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-31 08:01:09.000000000","message":"Patch Set 24:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":24},{"id":"5080f95deec8e9dc1ef3f4697307389b1ac5a631","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 08:05:07.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a52951e0ecb04cecb77e0acb61ed2886\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/524525ba0d0b43c09ad9ca98cb1c3bce : FAILURE in 19m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c556575f4ad4bfabbf3999ca8830e9b : FAILURE in 19m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/511c339e24bd4f8f88a0cd9e350d6529 : FAILURE in 25m 16s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"b6699810cb5cdf5fc4a1f8837fc4936b39939906","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 09:45:20.000000000","message":"Patch Set 25: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbdf36e6a2734c66a5486754d7075cf8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f92b12c6ae543fa938b82891862c108 : FAILURE in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef62bd0a09f1499ba06944c35a167c76 : FAILURE in 8m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df476fb37ae540cc87c1676401cfdfc3 : FAILURE in 15m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69a354979d294fd8b4c59b807d93573b : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/91abfe756b8a4ada90c3f88c05aaab8d : SUCCESS in 1h 06m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/efacf8457c7d4b558086ae79b47f5638 : SUCCESS in 34m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e686eb929aaa40bd9d7feca736598533 : SUCCESS in 1h 42m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc9bd4b03f59418dab8e5248dce2ee19 : SUCCESS in 37m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76aea1ce362a4cb3a050129329644856 : SUCCESS in 15m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0d484ed74307426980b9030cf48a9df0 : FAILURE in 16m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/75996908a6e3419caf92cd958e952c84 : SUCCESS in 5m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/58e0271dd2034181b5992b019b32e3cb : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f0e8383456a64b609c81fda7320a755c : SUCCESS in 6m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/32aa8eebd83c4a1abf516c55ec80e136 : SUCCESS in 9m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/150cb5b1b77a44159039e15697891482 : FAILURE in 32m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/764ba79265294b0495163f5cef80d36b : SUCCESS in 4m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6b7447f914424a5cacc44e113f8d6c06 : SUCCESS in 1h 13m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2751fa11ce3e4b439989c120c92e86e9 : FAILURE in 2h 28m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d372e7de39834155a08ba829398e1c03 : SUCCESS in 2h 11m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0b79edd9ce2c48108c9c27c7128d6104 : SUCCESS in 2h 23m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1300b6d8018d4e6199d4e8edca1cb65a : SUCCESS in 39m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0c1544e840da420dbebdae705087204c : SUCCESS in 1h 19m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c841b4e2587840c599a944405c4596a2 : SUCCESS in 1h 42m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a0b87da05c7745e3a7c490c8a6a679d8 : SUCCESS in 45m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7ca28863e63e4ea4964ad37714a55187 : SUCCESS in 2h 00m 52s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/02c08fc0511c461a96db95fe6f30aabb : SUCCESS in 1h 05m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16dcdff8929049f8a4194d31fe51c5af : SUCCESS in 1h 13m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65e5071d698747159c25ab0aacf74195 : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":25},{"id":"d6c4b76b50cb2f8c2a253eeac92855d757b6d6c3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-31 09:53:42.000000000","message":"Patch Set 25:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":25},{"id":"e3102591c3e743a7eeeb5c10eca3c69f79a6b4ef","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-31 12:59:34.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 13m 10s","accounts_in_message":[],"_revision_number":25},{"id":"ac0cfe873e9a6431b40f70ec4a39f473bbbac0c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-10-31 15:32:34.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"76d7714e81a33249c1f22c16ceacd5606602582d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-31 15:41:40.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":26},{"id":"0ad8a2fc9ba8bc4ef8d44948be94a39eb65f8c58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 17:09:52.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82ccf2f290da4c8da8e6d7b4b38e452d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c771485bd7424e6280026461da5c2866 : FAILURE in 12m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4aa74621ffef4b31b8550e9f8f540e2d : FAILURE in 12m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3de1c1564e740ce90759fa7d2a12502 : FAILURE in 27m 19s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"9aed0c089d14c5fde0e86791c4c7c193ec5f4997","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-31 17:26:28.000000000","message":"Patch Set 26:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":26},{"id":"4ee9fae86a8ffddb541000e296994fd29df17d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 18:15:04.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b9e8ccaf8244b049c981939c65baf05\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b01bdc2fd941d3be5a2ccc5b759c9c : FAILURE in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93fe7990964b42b0bf9ea0f9ba393282 : FAILURE in 11m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a09d620857f4f13b185aed0be43b5de : FAILURE in 15m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4af292b42344f9496b0be0367382f3d : SUCCESS in 15m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/4cae7ee8630546f78f20b7de7565bc61 : SUCCESS in 50m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea0a828485f14b9491effd2150e46993 : SUCCESS in 37m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1bf37a0e12ae43c581d9ae28e573733f : SUCCESS in 1h 48m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cc30d316db3443db48b46266e937fcc : SUCCESS in 1h 01m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09686dd8716c47e49eed67b77950333c : SUCCESS in 16m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7ab6c84679834d498365618570148e78 : FAILURE in 22m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/690fa432c1b2404d81ef4915f8af18cc : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff6ad42b653d437bb7bf2950b4d181be : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/daade15834614489a524c349a4f0df20 : SUCCESS in 5m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/da571275d74041e2b227cd97a7b078ab : SUCCESS in 8m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/274603dd06b64df6a5688db4c37af047 : FAILURE in 39m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9a5ccb57823a4ad98fd6db16d4f16e4e : SUCCESS in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/955e8c5e5bb54f3b9cd235f020a29c47 : SUCCESS in 1h 47m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ad76b1aa2f2d40d89d10eafcd17cab2c : FAILURE in 2h 28m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ad7637ca1093460d8bdaa4e02f08c99d : SUCCESS in 1h 42m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/dbd0c22ecd0e4cedb783465b5ce14b89 : SUCCESS in 1h 56m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/57c325764ff34112b95195cf2274449c : SUCCESS in 27m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d36ed948e5de4acebf7129fea9a96388 : SUCCESS in 1h 19m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7ed2f787e08d4b6997a5d189174db890 : SUCCESS in 1h 50m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/06f0332dc54646a99b0d430745027585 : SUCCESS in 46m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/46a9462e47dd4666829bfd9cbff0a271 : SUCCESS in 2h 15m 38s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/249ffa0943da45968c327b6a1e56e5be : SUCCESS in 1h 16m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d9698e9867cd41c69acd2ed86273720f : SUCCESS in 2h 13m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2bd82c765bc444ebd30d2b2ee7f12f8 : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":26},{"id":"befa8dc93d20dbdca2689359281008000f59cb3a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-01 00:37:17.000000000","message":"Patch Set 26:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2254/.","accounts_in_message":[],"_revision_number":26},{"id":"d9bcd61a7bd15541e343effc8c283e0190075f58","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-01 00:50:41.000000000","message":"Patch Set 26:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2368/.","accounts_in_message":[],"_revision_number":26},{"id":"b6ec8481068c91e0813efac23d099e5fcdd32156","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-01 03:16:53.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 13m 21s","accounts_in_message":[],"_revision_number":26},{"id":"a1318ab6cbabf08901242c4c2f3d8399e4d99f77","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-01 11:10:17.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/91c4fb948bed4992bd4903d2e18ccc6b : SUCCESS in 46m 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b9c0941701ec470b83a5dd66be81aa9c : FAILURE in 59m 01s","accounts_in_message":[],"_revision_number":26},{"id":"a42abd5040e3b168d5f2f3c03e0aa8e811fdee5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-11-11 10:20:43.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"4c4e941f3db4b9e880d8f648d3a9d643790fb948","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-11 10:28:35.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":27},{"id":"5a40fc6ba45ba5999a2385f9cee76069a10487f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 11:37:35.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed0ecbe67f9a4516a976c7399a0b3727\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fd558316354c461394e5315eb8bb6369 : SUCCESS in 20m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57b2d2f7cac24200b412379dc6d87b9f : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5e35d4e43cd54e5d8ca4e5162b403944 : SUCCESS in 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"8059f8854e14ffa15d637adddd8af6d586724127","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-11 12:22:41.000000000","message":"Patch Set 27:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":27},{"id":"f8d0c1f01fb2f8b3bc0e529ac1c3c3f06029a259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 13:12:54.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0d52a83117348eda2301988a34c8cd3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/589ae3011f49416d9f80f7f0dbc22f46 : FAILURE in 8m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/963a799f779b4bb69a8584e4bb627924 : SUCCESS in 12m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb04c464f8694cf29ae5b368cd3e2ec7 : SUCCESS in 14m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/988fae38fcd3420b97335008a2ea31c6 : SUCCESS in 15m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/9e91b50099174f34a415f1cb8e4de822 : SUCCESS in 1h 07m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/20d4838196bb4d9f9a5dc58bde62241e : SUCCESS in 1h 01m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c2ca60725e9499da9c48fc955079878 : SUCCESS in 1h 45m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a060fbdd9af486696cf59181b5aee1c : SUCCESS in 1h 04m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05e3a889d2ce4197b2bd4b597345d1a3 : SUCCESS in 16m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/44de8a7fc7e345c1908d4ae813df85b4 : SUCCESS in 24m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/44320eba40b4432da716f050473120bc : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/014eb52d3dcc496a92e8720eecfe20a3 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f63b075f37c7485fbbaa760099979bc9 : SUCCESS in 6m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/94b23440eef04de28ed2c27a6859fef0 : SUCCESS in 8m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/33eae1e221024d31a6a7cdec7dfe4049 : FAILURE in 47m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e88c168bbc145f3910549964919cc3c : SUCCESS in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7769654f001d45d9881ffb5a1db96375 : SUCCESS in 1h 50m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/037d219325ea48eba781844753e683fc : FAILURE in 2h 33m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/32447b979dc94e80a263ec65d52ae812 : SUCCESS in 1h 44m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3b1e90622610457ab1cce14a7f04f397 : SUCCESS in 2h 05m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fa04e44e73314c11b413b90b752a1c79 : SUCCESS in 33m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/19d9e4b8eb2d4822ba0c7d0fe9606212 : SUCCESS in 1h 12m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0df72a8532784d9c98921f2f463869af : SUCCESS in 1h 34m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/19576bfacd1146808201ce5c5dabdf55 : FAILURE in 18m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/29d8fc5cbdfd48b480d285fbfec9daf6 : SUCCESS in 2h 43m 09s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e8edc46fe064486d94602515531a6fe3 : FAILURE in 2h 08m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e53df6023dc14b5c90e5a9b897f79074 : SUCCESS in 1h 38m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db1707dfe1da4441afeb7e54c1a0e30c : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":27},{"id":"2bfd90531e333befb16a26e8a85fdd461e8a9261","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 14:37:59.000000000","message":"Patch Set 27:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2510/.","accounts_in_message":[],"_revision_number":27},{"id":"eaaa30f86fbad885e80a10269eec478257e517c3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 15:05:11.000000000","message":"Patch Set 27:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2396/.","accounts_in_message":[],"_revision_number":27},{"id":"187f4d0ee7c1c90fb1c902425f5b9996978cf96c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-11 21:27:24.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f0d253a287d34930b9dec43a4af7e5f7 : SUCCESS in 45m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/756f59e50dc74f12a220416e5ddd4fec : FAILURE in 1h 04m 28s","accounts_in_message":[],"_revision_number":27},{"id":"88dfe89c29bee80003fb49a7625f6d305d933a8f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-12 00:03:13.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 12m 32s","accounts_in_message":[],"_revision_number":27},{"id":"fd0dff9db986f848d2101194bcd9617b3b57004c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-12 09:35:16.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"57eb0a9fc551e0af101666f245cebadb686515f7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 09:38:02.000000000","message":"Patch Set 28:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2532/.","accounts_in_message":[],"_revision_number":28},{"id":"1e83b5b5585097891e7d33730cc5ed61d9ce133b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 09:38:31.000000000","message":"Patch Set 28:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2418/.","accounts_in_message":[],"_revision_number":28},{"id":"2a90c9d9fd4f6aadd3a1155d9bbae2989045f39e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-12 09:42:59.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":28},{"id":"5b74c0c5aa72d0a64ca0306c0f971c4432131f17","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-12 10:27:36.000000000","message":"Patch Set 28:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":28},{"id":"bdfe994bd6b0a2db82ea783afb772b7cd9ad296c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 10:39:22.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68d9c25d66644f31a2a35ce5896aabea\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b66f05bcff184358a56ad3cf00756bc4 : SUCCESS in 22m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a0552f9a8b3e41678bf66cbc859642db : SUCCESS in 21m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/70a480eb4f444d1884c40aff8b36c90c : SUCCESS in 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"097ba0cc7520e9d74323bdb0219df2f08b1bb95b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 12:46:07.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24d70b2008fa4dd680313de49df5cb06\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3f26e1853ec4f6ba7f699764c065f94 : FAILURE in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c00b41cf9e844279a46a97a81f7fb15 : SUCCESS in 11m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d0ce635f89e48a9ab696c1ca6c7cef0 : FAILURE in 27m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb7f145ba4445b5bdb4151d1f21ad08 : SUCCESS in 13m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/d55d9417db8a4d74bec15c92caaaa364 : SUCCESS in 1h 03m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a5520333847a48c3af43e74d38c823fa : SUCCESS in 1h 01m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c187918e8630458e927723a28edde9e9 : SUCCESS in 1h 38m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a751b5f20b14bbc8b47a1daae2f29b3 : SUCCESS in 1h 06m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea3cdc54cef4428598d96ebc71b0b382 : FAILURE in 3m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/87ba440738f0425a9ed23fd7d9711c63 : SUCCESS in 21m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/adf2b71b5da746ee800b15e7e44fa83b : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/19c1581bcf814279b9552bafcda51304 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fe10a25d58804adabfd23cbabbc09a97 : SUCCESS in 6m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d6a52eb3660d432d950f0d48ac439017 : SUCCESS in 8m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e7208bac8b36487e8df56ed8b6404fa3 : FAILURE in 33m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/85375691367f4e27a6705ffef641204c : SUCCESS in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5c42d1e6fbe942b196148e19498e3b91 : SUCCESS in 1h 52m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b6daf05d58ec426bb7300814ac68b457 : TIMED_OUT in 3h 05m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/959a8e829c244b98a084d03d9bcdfffe : SUCCESS in 2h 38m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8d639b0f8a174bf59d4aff119b8a3c11 : SUCCESS in 1h 53m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1503815f8aa2405fb077f3aca9af1ffb : SUCCESS in 37m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/48d52204425c49f0a2b4144eff0cf793 : SUCCESS in 1h 12m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e2081f0dcc2c478a84e3fdedd0bfc31c : SUCCESS in 1h 20m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bdf12f038ed14f71a568e5e265ce4e0c : SUCCESS in 43m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e316fbabe94e43a8bcd5432dc2bd5cd9 : SUCCESS in 1h 57m 40s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/89203362529a43a1986cf2d48cff21c9 : FAILURE in 1h 47m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4fb95d296c2e4b76be1c5cc0eb412276 : SUCCESS in 1h 47m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b477a3c2f84c4e44a92539855de61e91 : SUCCESS in 34m 58s","accounts_in_message":[],"_revision_number":28},{"id":"155d6e4860ed9e21c3c3fe92cd464521c5eb5854","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-12 13:11:35.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"bfc919f7f7148121f23e100c08d89c488d495ded","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-12 13:13:06.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":29},{"id":"56420bc7d6a1c55f7a29801a115312697dc17f54","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-12 14:05:25.000000000","message":"Patch Set 29:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":29},{"id":"18c7749e732c2c772334b5ff45523f50ff2a702f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 14:17:29.000000000","message":"Patch Set 29:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2533/.","accounts_in_message":[],"_revision_number":29},{"id":"f21faac778e9d91ea0f760e2581d2497a8bfef31","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 14:36:05.000000000","message":"Patch Set 29:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2419/.","accounts_in_message":[],"_revision_number":29},{"id":"3e341811aee4dc6bd3de956cdd72f465fd69a001","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 14:59:53.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9fb87ea1ef9411f873584f67b2f8949\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0aa731baf514493383484954fdebdbab : POST_FAILURE in 23m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f96d211612b44dbea19afe8a3ec846b6 : SUCCESS in 21m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/df00567f7c844892be83cb592dac5439 : SUCCESS in 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"898c18bb3bd4f1a5c787f3e17f38aca39cb8321e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 17:41:19.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49af6f584a254e6bbd708d49819adeb5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ecc89833eee4b3dbb7b6e46127d15d5 : SUCCESS in 7m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bca382b17a34fd6b258198f334a2328 : SUCCESS in 9m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a616e33c76814a4bb6f6f181ced12139 : SUCCESS in 14m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd1c52c19634636842a70410f9bc8e6 : SUCCESS in 14m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/d6c661b931504a1799254e3d1cec5b7d : SUCCESS in 54m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62481b8ee3cd4cddbc79e90e4ee8de76 : SUCCESS in 56m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba6c82081125475c892c2c52c0c35231 : SUCCESS in 1h 38m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cd5a3a7b28744a2bb593d318007f45c : SUCCESS in 1h 07m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9747c4edfb0f411695d45718eae4816d : SUCCESS in 16m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7fad1ac23f9e432e91e047efcc465b55 : POST_FAILURE in 27m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a1b89815c8b74ab5b6ca32de766202f4 : SUCCESS in 6m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/13c097a1092944ddbe3137a9c6e4a7f0 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1edbcdc6f4ea4e67b2cdebe0592ffa20 : SUCCESS in 7m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/491d3a8862c043878f5547bc8c2e6816 : SUCCESS in 11m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2e494175767a4192b26ae213b8a27c7b : POST_FAILURE in 40m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b17775acba8245c5969d1c133cb7f5ee : SUCCESS in 6m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1911de147563444b858b5843af063966 : SUCCESS in 1h 58m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6e3f1c572212497b9ee3da1b8f53e1d7 : FAILURE in 2h 39m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d214416520c84295b8517235650b39d6 : SUCCESS in 1h 54m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1f56dd3d0ad64442a13f0942cd4150c8 : SUCCESS in 2h 01m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0fb5dce4dea840dcba26eba59e95fd85 : SUCCESS in 39m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/11ef1b9b1c78436986024b11ec81856e : SUCCESS in 1h 11m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/731b1b6940074da9b23dad334ea4e207 : SUCCESS in 1h 41m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/42a41d4e45be4d6a95a4b6eb88d63d86 : SUCCESS in 43m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9b69a977765e4745b11b7ed5e5927e41 : SUCCESS in 2h 33m 01s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/8fe743c385a84ce7911fa6da70bf83f8 : SUCCESS in 2h 01m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b5d3ad477834d309485cb04668defe1 : SUCCESS in 59m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f577cc47d354574ab96c52f85863aa3 : POST_FAILURE in 1h 00m 03s","accounts_in_message":[],"_revision_number":29},{"id":"5c420288c09e382991c0276eb08f57708b4525aa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-12 19:31:18.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 10m 58s","accounts_in_message":[],"_revision_number":29},{"id":"563150a38a3f76572d2a8e385ccdc3658a2383a3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-12 22:58:24.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0f169604b30b414b89082ca7007b7286 : SUCCESS in 46m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c8e7d9aa201a48faa853eeeab9fb71dd : FAILURE in 52m 39s","accounts_in_message":[],"_revision_number":29},{"id":"c5feaf829448d88d7e9375d441b35ae49abd0925","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-13 02:14:45.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"ebd46edc5fb5d8a05e360aaf7c34f29a7aedd1db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 03:35:48.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327e0ab06e7649c7ad3b7564d7f29167\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6454b2d9a6ac44a482af11186cb5f151 : SUCCESS in 20m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c68c1f8b6f4448d88f86c98e364d02ba : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/27c92cb3174f4c1cb742a0337caca90d : SUCCESS in 17m 05s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"b8348e63683e400f53a1e2617436b946d9378e0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 05:03:32.000000000","message":"Patch Set 29:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a34887e368f4ffc95b35fd52f4514c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfa25b0f9a8646199023b2d37b122698 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/057c23781f944b98a1177d82c2f0403f : SUCCESS in 11m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fe9e158df02472c9d0b75798d23f216 : SUCCESS in 15m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef464918ad8a4d9aa90d7ac48a9813da : SUCCESS in 16m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/1fda8a9502db406b84d57b6551a3333f : SUCCESS in 37m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99d61ea1c9314974bcebe2d6be65907c : FAILURE in 5m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/179a7cae808b42a98e1ddd052ef5e164 : SUCCESS in 1h 31m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ade810397ec44263bbe8dd1b7cc0f3de : SUCCESS in 1h 02m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e9a665080154defbc75b9f8e27c722b : SUCCESS in 17m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b65a4d2d6e694662b3caeff3c330feb5 : FAILURE in 26m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5b79f9ebca6d49eab5481a9360103075 : SUCCESS in 7m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/882fdd0520fe46fda133ba8237968dd5 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10df0e0452f8433c82b43051756883ab : SUCCESS in 6m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4ff63c2f1e7c4818a21a09f3e06c42bf : SUCCESS in 9m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ae80db980691418fba56227798329a1f : FAILURE in 40m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/acd100c715a642bdb577e8c70ba2b2e5 : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f494d807b0a74000ad0fb7ff1515ff66 : SUCCESS in 1h 51m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d375b428e48a4a17acf4642ef6a2f144 : FAILURE in 2h 33m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fce77423388a4b999befdfc33beb2523 : SUCCESS in 1h 45m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/04dae882814c4adda9cb1b54d4917123 : SUCCESS in 2h 06m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a5fcfce26d52456d8c6ea0ff475d16b2 : FAILURE in 5m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ef396a5d1e8843929730589091d89718 : SUCCESS in 1h 07m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2d2d8cbf9f11442f8f960609779e460e : SUCCESS in 1h 27m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3b858314307e486a8ff70bc79f02967a : FAILURE in 5m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0d9a77eb5d6e49fbbc537da9e4fc8d43 : SUCCESS in 1h 44m 01s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/668d8e89e4d948c2b61fe52b13354e2f : SUCCESS in 2h 06m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/03b19f80173e4424a8b9f6e6d29eebe1 : SUCCESS in 1h 52m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8f65602847f4884a539f943776b1c04 : FAILURE in 6m 21s","accounts_in_message":[],"_revision_number":29},{"id":"d8518703d866a8b00ea8bb346bc33aecc321e004","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-13 06:15:32.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1df277bde47b4be0b1a43bb89e389024 : FAILURE in 16m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7f0ad5becd1f4c39b9deb05dfb071f4f : FAILURE in 14m 54s","accounts_in_message":[],"_revision_number":29},{"id":"1a2029202412b764af8ac70abd879482dd7cfd56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-11-13 10:22:02.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"abef9c3c81e6346b192dd01bfc5109d0d464479d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 10:30:24.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":30},{"id":"58c5e538e5d0bcc889a7baf274526f11cdc4283e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 11:49:12.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1b2f92d303a45e6b44aa45414edbc93\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/90e5a032c62d4055a3129f755c6d0993 : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8b6e1180f20d47f3be457a3acb4a6bcd : SUCCESS in 22m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7b55b742b5d046faa139b104d9d1ea34 : SUCCESS in 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"cd1b183fea04beef7376b99bfa2915450892d12b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-13 11:52:09.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.","accounts_in_message":[],"_revision_number":31},{"id":"0a2d48a1781ba6a01cc370a14a361fdfa0cd4094","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-11-13 11:57:28.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.","accounts_in_message":[],"_revision_number":32},{"id":"4b4fe054b01e912f0733b3521ee2e94390e2ffde","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 12:04:47.000000000","message":"Patch Set 31:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":31},{"id":"8c7707be36ad60489274406ff076e2a1b53242b0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-13 12:15:54.000000000","message":"Patch Set 30:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":30},{"id":"af88fd431bb00bbc4b302be22b08d83218af526b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 12:19:55.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":32},{"id":"fd0d0e6b59a3d2764dc18aa46b3fa8423f3d234d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-13 12:22:25.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-83-949483-29-fc : SUCCESS in 59m 10s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-83-949483-29-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":29},{"id":"f8851617a26745df2f7eae84fe8079fee4232cf8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-13 13:50:47.000000000","message":"Patch Set 31:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":31},{"id":"e9858090bad799ada81336b7d8500fb309f9e187","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:03:37.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/437d6cbcb09140eebddedff426ddef1e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b818a361a984920bd178f4b7f825014 : SUCCESS in 21m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/270686739b5d4a059a45f0114d543a32 : SUCCESS in 22m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/48aa3ee810874430ae15f018e0e4584f : SUCCESS in 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"51cb4a6e00ac3f0936fe908fc66e517a10637e06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:18:27.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d216ed602f444e2acd53b27488a566f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03214d9796bd466799b7c99cb6d2b214 : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9ad50c31e4840349e161e1def3977df : SUCCESS in 13m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b2620b4bdd54c089d16704c1bbbe406 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e9806e9eb4493f89c92a2a9c9104fb : SUCCESS in 14m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/bc6da052ed194ef3b9400f8e73d00f5a : SUCCESS in 1h 04m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73b102942029450095d13a7f82c19fdd : SUCCESS in 30m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a15a223b00245ee96845e01af8cf90e : SUCCESS in 1h 37m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b3fe266c0714bb990dde1b6f42c532a : SUCCESS in 1h 07m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/023e3e54825542a1a8baab51eb6d1a27 : SUCCESS in 16m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/731895e3df034305b0634d1179bd7743 : TIMED_OUT in 41m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/afec6360c4734176a7d0fea5debc4d71 : SUCCESS in 6m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/38454fa71c6b4b06b6af5251e9597e2e : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b14244d3210e44acb6e4a62fcbfc5990 : SUCCESS in 6m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ba44b4e37af843989b9fa0599d735c01 : SUCCESS in 9m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c17b31684ace4995b97579304fffcb10 : FAILURE in 40m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a37c3247ca834b9dab62a3086e38745a : SUCCESS in 8m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ee880848a7b14c3b9288fbe47893a386 : SUCCESS in 1h 58m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ad24b16f6f4445ce9d2a798f10cdeef2 : FAILURE in 2h 53m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1ff41db8617c4d678d7b3a3d82dfbd5d : TIMED_OUT in 3h 08m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f351f6b2ef7b4f43978d09489cb4c663 : SUCCESS in 1h 48m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/834fea7aa39a423a8fa6dfb97d7e4e53 : SUCCESS in 36m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aa0401a1530a47d399f77c693608405e : SUCCESS in 1h 00m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2cacdbe49d1647799d69ed814486a8af : SUCCESS in 1h 31m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/195fa7bc801341efab22696721861ed6 : SUCCESS in 34m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b2ba5f99654745cf9d2c9e1370443c18 : SUCCESS in 2h 09m 35s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/4720bd1db0f845d0af36b9d5a28e0456 : SUCCESS in 2h 08m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/04f48380ff5947a8ae6d621270e8567b : SUCCESS in 1h 46m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f913a1c509544ee19601803920a995f2 : SUCCESS in 33m 28s","accounts_in_message":[],"_revision_number":32},{"id":"1104cfae6a8366c2efce6e07f3f0db45d32b537e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-13 15:57:09.000000000","message":"Patch Set 32:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":32},{"id":"e0eeda9c8095b292126015b79024b3162e164dbf","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-13 17:42:27.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"d3a90cb1563208dca058f9cfe13d920e0d74a223","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 18:16:31.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6a1eb3427ae45d793a0faad127bcbc7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9538138d3d064080b776fda0ece3b9d3 : SUCCESS in 26m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cd7f5db7abca4e36b7476a1f82053beb : SUCCESS in 26m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dd4916b544c24dea8bd04caf5b13514b : SUCCESS in 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"493f550f77edc5fccaf75dbf855c01356fab1299","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-13 20:04:58.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6b913ac0d12a4fdea717398136f78784 : SUCCESS in 50m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/825ce1540a91479981f1d2e05c742910 : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":32},{"id":"4dd4c003c6d1e19669cbaf70debb0de885fac926","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-13 20:28:42.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 12m 38s","accounts_in_message":[],"_revision_number":32},{"id":"78db3e4640734ef1e03bd9461daf702074765edb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 20:28:50.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7054e32788a848bab0c23f96d4e38c3f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/888b48e69f524c009c69b4854e81485c : SUCCESS in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70be73c66e564ef3adf702c357a0d6a1 : SUCCESS in 9m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33f4fb10f1dc4725badacc212ede3917 : SUCCESS in 15m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8aa557505ca47dfb2c20145034047fa : SUCCESS in 11m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/2947150c0f6848eb972b3bcea67bc9f7 : SUCCESS in 1h 04m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/41731f6b97af47f1ba3feb77999e5d62 : SUCCESS in 48m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b3e90d4912034946a7b60d635357e01e : SUCCESS in 1h 30m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc08c1337cbf4f23bca1064d90e3ea0c : SUCCESS in 1h 02m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eda41fd0197b4f29a740d3f8c37b2c28 : SUCCESS in 16m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3474286dfb4644739a94c1e4fc44218b : SUCCESS in 16m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/afba2083d7bf4715803431a668d50d28 : SUCCESS in 5m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77e1c19d11f649989087c6c274c0000e : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1943aed1afe14adb9bcb12185b207233 : SUCCESS in 5m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6d04b029f04e45f587b5d7b5db261355 : SUCCESS in 8m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc21f5208f4d4c81bb8c85123daca712 : FAILURE in 22m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0c7c559fb3ea4ce9b5bfee8e93c7fb63 : SUCCESS in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/818ef7eb7ae244e1a1d5cff805db762c : SUCCESS in 1h 53m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a8c614cdb6d8459cb31c96cfec6a40e5 : FAILURE in 2h 39m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8cec023803974f74990c2041518adc58 : SUCCESS in 1h 58m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4ad7045215804d1d8d6050504b8bfade : SUCCESS in 1h 51m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/89d6837f385d418fae79427e60a10291 : SUCCESS in 16m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1384a843c5574dfcaa1f5bde6c70dce8 : SUCCESS in 1h 05m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/87713c0a69604869a85fe32c2505244b : SUCCESS in 1h 32m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9186290537a347a69d7d42f5e7d4d67d : SUCCESS in 36m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/38c794c676f749bd8ca444f5438907fb : SUCCESS in 1h 06m 07s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/89e7bd1c7362456b8e953d858183a664 : POST_FAILURE in 2h 38m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb60b8735dc940918c9658b06c3c4a74 : SUCCESS in 1h 36m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a90040db00df45cfa8864acbdde32f7f : SUCCESS in 30m 27s","accounts_in_message":[],"_revision_number":32},{"id":"2afcd3686e664d5be973429d06978549121aa709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-11-20 05:58:07.000000000","message":"Patch Set 33: Patch Set 32 was rebased on behalf of \u003cGERRIT_ACCOUNT_37470\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"_revision_number":33},{"id":"3ae8b0965d7130c39c074bbca8e3d67b299484a6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 05:58:22.000000000","message":"Patch Set 33:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2531/.","accounts_in_message":[],"_revision_number":33},{"id":"c0fd020a0afd598089367ad11b60681f66db9773","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-20 07:03:19.000000000","message":"Patch Set 33:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":33},{"id":"44b8bcb56997f4560cd470388bddaf996d515d94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 07:26:46.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/805f5880e5914b6ebbc6dbdae6922044\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b08284b676df498aa0300e198850a132 : SUCCESS in 25m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e87bc64f29f4ac2b6ef84f9f74c6d0d : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05b75e82f9894008b0819674415bc5b4 : SUCCESS in 33m 11s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"fbca8d8d93014f6b3cf999cbbaaea102c5263f7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 08:34:34.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f51be509431c426eadcd85f0b6f6bdf1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef23b2c6a08e4acc81b6c431ebad965c : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1abeb084fdad4e7aa6128601d47e057a : SUCCESS in 11m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d73cee5458e847a4838f9a04be608d92 : SUCCESS in 15m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3160ca6ad824cf9a25f94d7a3118301 : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/42b43773f7864d29bac249c64e794214 : SUCCESS in 1h 01m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/39c506e6a05b48798d3a15c4c13284be : SUCCESS in 56m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/78c9ad4316714a82a53adedb4cc62fba : SUCCESS in 1h 37m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a7a9c1e101d4ce78122334cc6714372 : SUCCESS in 48m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/562a797253314f4198875c55fbae9583 : SUCCESS in 16m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/33edac80094641fe93c317409ece7f0e : FAILURE in 25m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e9b9f4977485403db1019e4f6c3f2a07 : SUCCESS in 7m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/093e92235569421ab9130d218311b662 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f65e23d0009148c097c78b4c21adba74 : SUCCESS in 6m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ec3e929571ae4787a1b4ef645ad9a8f1 : SUCCESS in 9m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0a0df9a64a45468b89f5eafed6b00d63 : FAILURE in 26m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/26b13c646166460daa64f1dd182230c1 : SUCCESS in 6m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a0141fdf83c4a54a01aeb69506fa58d : SUCCESS in 2h 00m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7bc4ed71623245cb9f67e2a95d7c5cbb : FAILURE in 2h 31m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aa74444f13f3496183592bc0c21b55eb : SUCCESS in 1h 59m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/43c6c8d8a4244a07afdfba03e85a649e : SUCCESS in 2h 17m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/99d50e8c6ba846f0b26a766f42fc5fbb : SUCCESS in 36m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/17ffa4c5432f489cb98e18da115435d7 : SUCCESS in 1h 06m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/83463be130e84e069ef486dc3838d16a : SUCCESS in 1h 52m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1cbe1eb2180647fcb56b7e405cbec2e0 : SUCCESS in 39m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/185c82a094104fc1a614ff4349867d69 : SUCCESS in 1h 53m 35s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bd7180e275c9464b85c6fd616b0a188e : FAILURE in 2h 01m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aee0295d11b84579aa8566a952ef41c9 : SUCCESS in 1h 45m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/344820aa14fb4a8a8746b0352b0044bb : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":33},{"id":"e4884e617cf20d4a33665266216cea44e082fefc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-20 08:47:57.000000000","message":"Patch Set 33:\n\nBuild Successful \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/30/ : SUCCESS","accounts_in_message":[],"_revision_number":33},{"id":"0926cd67dcaa84fab6456ec0d6805fce0b0f0018","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-11-20 08:48:14.000000000","message":"Uploaded patch set 34: New patch set was added with same tree, parent tree, and commit message as Patch Set 33.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"6155d075c5c9131ec1a2e2553332d6ca2e0cef38","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 08:48:59.000000000","message":"Patch Set 34:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2538/.","accounts_in_message":[],"_revision_number":34},{"id":"df3fb13fc6c0117cf708277072e6f2389c28e02f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-20 09:07:13.000000000","message":"Patch Set 34:\n\nBuild Successful \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/35/ : SUCCESS","accounts_in_message":[],"_revision_number":34},{"id":"e2ba8c9f4383e7c3a96da9c57cf2ab0574d9b1c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 09:31:39.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23ce9c14c2fc452d8a77ecc733555aa1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6121fdb273a4bf3b09cb8a672d80103 : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce3cb46586de4b0fac00b7dbf257132b : FAILURE in 29m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ec487df9762146aa90b22c46ad7b9b4e : NODE_FAILURE Node(set) request 3165039725ab479ab52b5d3a849a5e94 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"936e80c1fd2ce7fbfc570e5d184c068429082e76","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-20 09:50:09.000000000","message":"Patch Set 34:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":34},{"id":"e0f7f2fc5bbed2ac126edc52a14984a270f4354b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 11:39:21.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3633b62ef7a4b4b8e8f16d2cd53b522\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a77525741fad49f2b331618e9ba1b47b : SUCCESS in 8m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9792d9c2943b4446b24750bdd2031d2c : SUCCESS in 9m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff94e0daa85f4380bee62efe55155912 : SUCCESS in 15m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf02fdee47b448dbf4183b0fcc130ac : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/6be9fcd568e542178507f21a847cc37b : SUCCESS in 31m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b43e3522d24e488b9510f14514d2d861 : SUCCESS in 30m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9fe07757df94995b8534ea1790e5c66 : SUCCESS in 1h 25m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab0fdc1502e24bc6b7e8dfefe8d89a0d : SUCCESS in 1h 01m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ae05bcd887e48a4b934ffc629c2c87d : SUCCESS in 15m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d2bba19828f34430bce9ed0458d20499 : TIMED_OUT in 41m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/756e54358fcb4948bf199a138a5774d6 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e4eb6f851f245c3a8729b75925abd46 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/38a2abcca76445ae89ea8e1606ccc67d : SUCCESS in 7m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/296d225a9e944e90a7d17ecd92f59163 : SUCCESS in 10m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d847c583c7c14d5c914220b65dd6924a : FAILURE in 40m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/924d9a75f3b84a17b5efc45ee2afd451 : SUCCESS in 6m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c493483d98c243ba9c9e6a4f6af93c42 : SUCCESS in 1h 52m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eb215f01ab004a8cbd136eff7271038f : FAILURE in 2h 39m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/90106eee9027413da4c0e582f05cfac9 : SUCCESS in 1h 56m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ef2da67d60f14f1aacb2fb0f17f5ff15 : SUCCESS in 2h 00m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/77974a82420742fdb99540c96927ef0d : SUCCESS in 37m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2e8ff3cefd804473b556413d24a7ab01 : SUCCESS in 1h 17m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/58ee55fe812c43afae0cc9143ed53ec1 : SUCCESS in 1h 51m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a019ee5b01a24c9bb070e3ac80914f84 : SUCCESS in 38m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/54a1548f050d40c2ba5cd82ed6a499f8 : SUCCESS in 2h 01m 58s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f551a50c30134cadb8f122d46291cd55 : SUCCESS in 2h 24m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b7ea0836b1934e038fff0c168ba4d32a : SUCCESS in 1h 15m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e7e9eebddb647edbc38dd36b5bd94a0 : SUCCESS in 1h 06m 17s","accounts_in_message":[],"_revision_number":34},{"id":"95057d96df832aef4881aa598f3b35c8863145ff","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-20 14:03:43.000000000","message":"Patch Set 34:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d81b336fb0144810abd212a3a698167b : SUCCESS in 52m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d420587960654b589c141a2d4560654b : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":34},{"id":"7ca30d5c4fd533c734f2752b692de2aa166217cf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-20 14:17:42.000000000","message":"Patch Set 34:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 11m 15s","accounts_in_message":[],"_revision_number":34},{"id":"35c5378743d0a3dc171792786df3a0c61badb6f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 15:51:47.000000000","message":"Patch Set 33:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2675/.","accounts_in_message":[],"_revision_number":33},{"id":"f6673da8b8dba58a09258b84d7fe2a83642b6a9f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 16:54:37.000000000","message":"Patch Set 34:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2682/.","accounts_in_message":[],"_revision_number":34},{"id":"aa5040a23c18a294fd761637419e06210a861a88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-01-05 08:30:10.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"4a7cc9f01f4bf6b7da1a986184ef8d4d8e225e16","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-05 08:51:07.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4320f8b917bf4c568a455da2eb1c36f7 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e1a8ba840dc141f7bf0b067c070a0509 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":35},{"id":"ef65d158466655259824d6109ad693ad6d00e173","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-05 08:53:30.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/f0dd53f3a5bb40468fd3b9f1f571dfbd : NODE_FAILURE Node request 200-0000942314 failed in 0s","accounts_in_message":[],"_revision_number":35},{"id":"8c6e88e149620fb0248285b53b1fd8202f96bebc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 09:35:07.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4752a3e8d0ad44949f2a0fc956e9ae90\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3ae77f6cf87402480b950689f94c757 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6d774a64c3de4ece9ca4937860da2ec1 : SUCCESS in 15m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f91aaa0bd57740768d1c0b696d0c37cc : SUCCESS in 15m 38s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"3251d425906e6b0e6439638076c3d4549216a1e4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-05 10:26:16.000000000","message":"Patch Set 35:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":35},{"id":"8f9980512c544fba43018d8071d97284e3188396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 11:16:01.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3291cc379ffe4f0586371165ead532e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/649f8b532bc642d4b10e5e612df4b308 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d417ff7255844dfa657172ab5f43fb3 : SUCCESS in 9m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/942ebf6cce0d45c49d663182ea086860 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fffa5c4e10154875a6f04ee3dc6cc9ce : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/dfb4e7ea7bfa40d39d990e679933cb20 : SUCCESS in 1h 03m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1e6eb4498eaf4c4cafe2212f4ce04c4a : SUCCESS in 54m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/512acc0ed62f4401a41c14ef0629a5ec : SUCCESS in 1h 19m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6b7865b98b34fa9b91d324aae63f45e : SUCCESS in 1h 00m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/835566cb4ac645979113b8a7c0959920 : SUCCESS in 16m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3d9b3afd5c994daebf5edf24b4f7f61b : TIMED_OUT in 41m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/18ea10045dd94e06af85c84612bab4db : SUCCESS in 3m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/17ae517fb44143999f4a392681e52364 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2db922535134ecfad8cefa49d1b7641 : SUCCESS in 5m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8438d889c9fb4af0bad3f5a532304c6f : SUCCESS in 10m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c8644558ff354419b0f0fe34a49d96b7 : FAILURE in 44m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/68470fb9fd844319b30b9aa8b13fe537 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/737e3b70094d46f680c0042149700f5a : SUCCESS in 1h 08m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/444ea6b44bdb4519a008f0533cd844f1 : FAILURE in 2h 28m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2badead457ad46caabe0258ca25c3de3 : SUCCESS in 2h 16m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/43ed1dd7fb154a66a561cfc6076dcf66 : SUCCESS in 2h 04m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0808cf45fb5b43e48c35f161a5c2e1d8 : SUCCESS in 17m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/afc91496ed464bdb95604c6f57bae994 : SUCCESS in 1h 05m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c1946642e2c6462f9da6c82ddddbd4ac : SUCCESS in 1h 26m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b110372098d44eeba7fd6784bc29adbb : SUCCESS in 46m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab973f3bdd1a4219aa4322be5705c55d : SUCCESS in 1h 13m 40s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2eec2106ce5e4e63881f6f07d7ee932f : SUCCESS in 1h 36m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/91f8573c78e04e2385e02e8f536d52c4 : SUCCESS in 55m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdb4a31e8f8a4855b3c88add3dc38d3a : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":35},{"id":"3d82f7bffffd5738a80ce47779aa8baaacb2e467","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-05 11:27:12.000000000","message":"Patch Set 35:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_01_2026_11_20_09\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":35},{"id":"5e350228f62eade68dca02e196ec2a235f444f1f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-05 13:32:55.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 19m 11s","accounts_in_message":[],"_revision_number":35},{"id":"e50f44b06b94c7cfdd85c336cb0433de6bb34c69","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-05 17:40:51.000000000","message":"Patch Set 35:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/66/ : ABORTED","accounts_in_message":[],"_revision_number":35},{"id":"2408a568c633c396a4e45fdf6fe1ecd25304865d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-01-06 10:23:31.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"55e03d5aa9e67e0ade4c75e5aafc07d16f8f9999","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-06 10:46:32.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6d8e9b529a3f41bd8ed2de2ad1847089 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/edb4cdcb4eec4009b30f2a45bd6bb511 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":36},{"id":"0255a1924490ad25bb9a9ad0598882cd0066872e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-06 10:47:54.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/f6ac480387d44141aecec0edab05f101 : NODE_FAILURE Node request 200-0000942337 failed in 0s","accounts_in_message":[],"_revision_number":36},{"id":"2db272595c67438aa96afc6b619d5364fb83347e","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-06 11:16:34.000000000","message":"Patch Set 36:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_91/index.html : FAILURE in 52m","accounts_in_message":[],"_revision_number":36},{"id":"eb90ec0ba45322e441ceafa4c9ea810c1ae2f150","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 11:16:52.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d35448597b7b4674befd4533ef63abc0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/88066892591b49758ca08dfaa400ae26 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1cce2bb4bfc645b6a51338a5467e15c1 : SUCCESS in 19m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/001b9f920a334f048ccd18dae2245eb6 : SUCCESS in 24m 06s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"3e1645c332b43d3a948ac6cec0e6121cbb2dc8bd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-06 12:34:24.000000000","message":"Patch Set 36:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":36},{"id":"edba6adf695b66ad017555ba79d8c78bccb53ee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 13:58:12.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77f5d5c0ab3449e6b1785f21955788be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1287d3714b12423b9cee3733f8650910 : SUCCESS in 10m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0096cc8eb1494528913767362b43fedd : SUCCESS in 9m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b7518f4e2c549feb70a628d35ff6f49 : SUCCESS in 14m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6f079b04d754d0cbbb83d721398c6f1 : SUCCESS in 17m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/f50c4f78230047d3a3182bbe6e4b1f2b : SUCCESS in 38m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a2b66613f0d045bcbb6ae2378ad5bff6 : SUCCESS in 58m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e3d31efbfce6471f947712ba597dface : SUCCESS in 1h 44m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ca776d4a5c74971aae26e49592d4363 : SUCCESS in 1h 41m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12b00a44b535482f9db8f0eea5208b14 : SUCCESS in 14m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4e4017b9be0948f0b56c9fd2e551f4b7 : FAILURE in 26m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0f69f5cefe3e47988a67ddb16e68da38 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e8e22d725689479e9a695840ef53326c : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b14ce6fcec647ef911dce18ab5d49e2 : SUCCESS in 5m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ec3f927d45a0410497649e7858995813 : SUCCESS in 18m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cd62dbcca01049e0836b341964e94e2c : FAILURE in 39m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cd86dbec0d9845fc8fda9c172c75f5fd : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b95793f87c8849d0b109de3a7fef5dc7 : SUCCESS in 1h 58m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/42e3f034f1fe4831bb33d4f753e8f12f : FAILURE in 2h 56m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1ee3580b3ea1475fb52052a07c846af8 : SUCCESS in 1h 55m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e8d3e55d9f20444d87e5a46ab9499a9e : SUCCESS in 2h 47m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/003646e0522b493185b354f3e25830fd : SUCCESS in 33m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/61e140232af9490988cd988cf3c2f4f7 : SUCCESS in 1h 08m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e6f2037f3ddc483b8512ffc42a01ff86 : TIMED_OUT in 3h 07m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/488ccebf391345eba561153424846909 : SUCCESS in 42m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5796d5d47f8e4921b2cfa57febd04fc8 : TIMED_OUT in 3h 09m 38s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e7dbe995bc0345d0bc61b3aef7701d92 : TIMED_OUT in 3h 09m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c0fa6c544de4af9aba42d767a5fb0c0 : SUCCESS in 1h 36m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0a07ce2afcc4d319e7db191bea9a5cb : SUCCESS in 1h 38m 16s","accounts_in_message":[],"_revision_number":36},{"id":"96b34dec45ee1f46fffd204295d05088d6e7c4fe","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-06 21:09:43.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 22m 47s","accounts_in_message":[],"_revision_number":36},{"id":"f265673b98df820ce1dac64af727dd4e932f2060","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-01-09 09:38:03.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"e7373b0ce5de23f92e5f99ce8f6cc85ffa9adc0a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-09 12:17:53.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f9e5555f67f147cd852ad1eddd08cecb : FAILURE in 1h 07m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c324532af15b4ef3a1088ac99a6fd3ee : RETRY_LIMIT in 15m 56s","accounts_in_message":[],"_revision_number":36},{"id":"83226a07b144253ffc2566ab1efc438efaeeb6a0","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-09 16:14:07.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-83-949483-36-fc : FAILURE in 1h 5m\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-83-949483-36-iscsi : FAILURE in 1h 10m 57s","accounts_in_message":[],"_revision_number":36},{"id":"103dbb03fd943ccb1ca058f1fe940767df00c6f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"real_author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-01-22 05:37:55.000000000","message":"Patch Set 37: Patch Set 36 was rebased on behalf of \u003cGERRIT_ACCOUNT_36538\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"}],"_revision_number":37},{"id":"964a832bac16edaac1ab1044535bc86412d2e80d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-22 05:40:09.000000000","message":"Patch Set 37:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/d61906f02978409fa191a52fd17f46ed : NODE_FAILURE Node request 200-0000623308 failed in 0s","accounts_in_message":[],"_revision_number":37},{"id":"ff7c4ad31e13f6775d0536bfb3ff5e4235ea77e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 06:03:41.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/581cdd13a7d84b7bb383869d7be02db7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0bebc35facc041dd8395fb3a81fe2345 : SUCCESS in 21m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e1df88400624af6adb6136eb2fe9476 : SUCCESS in 22m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dea95c03eece41a79c0859d18bc056e9 : RETRY_LIMIT in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"572d8f494e21d351563a534d54cda4077c475ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-01-22 06:04:03.000000000","message":"Uploaded patch set 38: New patch set was added with same tree, parent tree, and commit message as Patch Set 37.","accounts_in_message":[],"_revision_number":38},{"id":"6b213747e4f5e84b24199a562ea9e68ba6c5d537","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-22 06:07:30.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/c632fe4c54f044cb9f34623dd442a028 : NODE_FAILURE Node request 200-0000623310 failed in 0s","accounts_in_message":[],"_revision_number":38},{"id":"7ae650a15959e99c75fc8eb83401f217c27438c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 06:30:35.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed4ba6016ff946efb6f3c3db8dc37e14\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3f31d8fb76474d66b33383b600ce1c63 : SUCCESS in 23m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e612405e07f4648bce7c20cf89f2afd : SUCCESS in 23m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bdac027faa1f4a53a007d7573c074a27 : RETRY_LIMIT in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"c4b510e727bc821149258dacb48d8c6da1bcc31e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-22 06:37:36.000000000","message":"Patch Set 37:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":37},{"id":"827037d70666a16986a5892054f0147922250b62","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-22 07:10:12.000000000","message":"Patch Set 37:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_321/index.html : SUCCESS in 91m","accounts_in_message":[],"_revision_number":37},{"id":"d67de9d1544d6978a0844b05803bcfdaf6d6954e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-22 07:10:42.000000000","message":"Patch Set 38:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":38},{"id":"de2332f465feebdebd66ebd62c9695f20fe42121","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-22 07:31:31.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/d2f5a3265b9d468e9c86cdbd2cae1b67 : FAILURE in 38m 12s","accounts_in_message":[],"_revision_number":38},{"id":"db803c3a113730ebb3830a0d377f7a71f5d00b20","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-22 08:49:47.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 1h 18m 56s","accounts_in_message":[],"_revision_number":38},{"id":"08947c22cc6aa99269f69938865e47a651a18ea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 09:15:44.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/747759f1d4a7491eba202a44f9786dcf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aec3a0eea236403298febb8e7ddaff01 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff319f4c75cc47a59020ced868f5ee91 : SUCCESS in 11m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96faa725ec4149ad97c5b9f9788b2403 : FAILURE in 21m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef96c562932341f8923f519f3a5b8afa : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/46040c64a6a7452fbb353a59a33c9eb3 : SUCCESS in 38m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9983190ef2f4f4c87f9925c9a1bdb02 : SUCCESS in 59m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f66f44e5c9704a048d77815367bf6426 : SUCCESS in 1h 35m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66bfa09e35de49a18942f7cc57b185cd : SUCCESS in 53m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27431436958040cface5dbcc6c3ce60d : SUCCESS in 16m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/664ced9925c94998b64d93af8c77d6fb : SUCCESS in 22m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f336e90f64c24199b33d1815f28af2be : SUCCESS in 6m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf8a39dc3aed449ba34bbca277bb6b44 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fd76709c6c4e4e0db3abb81360809a95 : POST_FAILURE in 6m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bb2275beb5f54223bd40ae6526618f8c : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc3b1e4205c446ecb96f8751d3cd4f10 : FAILURE in 39m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f745f7f37df44b38969196577a79d3a1 : SUCCESS in 7m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/15c13ec3a0924ef09df651fddec5d4fe : SUCCESS in 2h 05m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fcc826ee25f44cf4ac988be49be1eba1 : TIMED_OUT in 3h 05m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/204f9fd2fc284ca998ff1e078f7ed837 : SUCCESS in 1h 42m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/606cf43bf5044cc28f9ec68df271dcb0 : FAILURE in 1h 32m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c81cae03ca444d32af98c2c361b48ce1 : SUCCESS in 24m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/922d60c25701426f9351c8bc607f36e5 : SUCCESS in 35m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d7f75ddc25ef4b749ea7e2ca3f7273e7 : SUCCESS in 1h 34m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7c5f4b5a177e4ffc83251de90fbd8f47 : SUCCESS in 32m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4ab302b5f104423ea72ce1fb5e40e675 : SUCCESS in 2h 19m 09s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/cfefe49ed0634fcc8103c0d889a9c41a : SUCCESS in 1h 56m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/78e46407a9ac47e081b4aee89223204a : SUCCESS in 1h 39m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b57cf4ec033e45a1981956f39bb0fcb1 : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":38},{"id":"08c4dbe4ba1b245c144ab78bd636c9d2a33d3483","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-22 10:10:10.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/932cddac45a44e6f91108c97ee5286b1 : FAILURE in 1h 13m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8d98aca3852847d296ef1dd228452bd4 : RETRY_LIMIT in 18m 30s","accounts_in_message":[],"_revision_number":38},{"id":"396f94d7c0c2eb875044bc55ba4b50912eb2e5c9","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-22 15:53:45.000000000","message":"Patch Set 38:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_323/index.html : SUCCESS in 101m","accounts_in_message":[],"_revision_number":38},{"id":"cff8715c2e74d1478bda895d33b66745a47f8d44","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-01-24 17:30:39.000000000","message":"Patch Set 38:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/38/20260124160929/NEC-ISCSI : FAILURE in 10h 00m 56s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/949483/38/20260124160929/NEC-FC : FAILURE in 10h 21m 07s","accounts_in_message":[],"_revision_number":38},{"id":"3e1cf4c5bdf193475870897e4a5d6c82e208147b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-25 21:04:38.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-83-949483-38-fc : FAILURE in 1h 22m 47s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-83-949483-38-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":38},{"id":"5f425bae5380a4c8f0ff52beecb8c5e75875da90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-01-28 06:59:48.000000000","message":"Uploaded patch set 39: Patch Set 38 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"b12b12bad2b9d2275c0e27b002340b3cc4921647","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-28 07:00:58.000000000","message":"Patch Set 39:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/689/","accounts_in_message":[],"_revision_number":39},{"id":"7b33515293164c94fec76d9d284ab57eb939020c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-28 07:00:59.000000000","message":"Patch Set 39:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/39/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/39/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":39},{"id":"99b4ca22f7f35b6e748c02b78bf21aa43ba5491a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-28 08:59:30.000000000","message":"Patch Set 39:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":39},{"id":"b5096d9601a461812dacf5aa8de2ff5ee1def528","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-28 09:34:40.000000000","message":"Patch Set 39:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/943364, https://review.opendev.org/926844, https://review.opendev.org/958465, https://review.opendev.org/951157, https://review.opendev.org/949006, https://review.opendev.org/929906, https://review.opendev.org/938771, https://review.opendev.org/916722 are needed.","accounts_in_message":[],"_revision_number":39},{"id":"e246010d0df86eb74af290ec0ca45276cf26f370","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 09:47:12.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4bc1df010c3463abb04a5bc36c9a8e5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618cae3f625046e782caf93e5df19463 : SUCCESS in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86c838b8aeb94191a5757bae220eb801 : SUCCESS in 12m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0199f294a41a48c1b561d7929f1dba11 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf6cf23833f4a78b9026f8f3d07f5b5 : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/63df9ea49f28486e8a5deca7840957aa : SUCCESS in 38m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/32f7fd9d332c4999a1a15093416f42b9 : SUCCESS in 33m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aee4619511c64d7a8adef48de73450aa : SUCCESS in 1h 14m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a41d90b18df46c0b541ba71d5d06f86 : SUCCESS in 36m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd919660e9b64369a459229106e8f8b4 : SUCCESS in 17m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fbc69c77daea40af8d9640a5362f0f1d : SUCCESS in 28m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c34a30fd2a484abca7cd53224de0e144 : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb48f24d2d314211ba8f59f386d29447 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8dafd0a1adbc4c6cb4124e096bf7908e : SUCCESS in 4m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/27b189fc2d13426e92bb32e3e3734f4e : SUCCESS in 5m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/314be836841d4652ad11bb52b891f376 : FAILURE in 30m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eb9b197c55f84ecd8dc003f7aa038096 : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2200ac4583f14469b1896b11c7157e2f : SUCCESS in 1h 21m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/287e83c0380a47bd82fe8ce8f5b952f7 : FAILURE in 2h 40m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eacfdea711344430ad43d9806d43270e : SUCCESS in 1h 40m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7a946588e37648f79a5cdc14639cf0b6 : SUCCESS in 2h 14m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/096149e491f744adbf58cb8692d39723 : SUCCESS in 21m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/54d47747d19c497a8fa840a78e4aa103 : SUCCESS in 39m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/06465b56a06e45208b969f870259d62e : SUCCESS in 1h 46m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8c117c05c94b42afbde59d6044700fcd : SUCCESS in 39m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c1305cf623944df3b2e5c638b4a447bd : SUCCESS in 2h 03m 27s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/6830f949e85e4c0b93739d96f663e665 : SUCCESS in 2h 15m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c04e8462c735478694a035ff254c2578 : SUCCESS in 1h 52m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09b58312d92e4bee9caecfebe4ae605f : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":39},{"id":"0807717cbe925b30ecd12ca2777b989448d34616","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-28 11:41:10.000000000","message":"Patch Set 39:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_380/index.html : SUCCESS in 92m","accounts_in_message":[],"_revision_number":39},{"id":"17ae5198f93e53e7e1a2721464cc44d528bfa3ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 12:09:14.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5da6b4a2639841f08b0e146c3cf6377f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/75ee9ef5ed2e482a9c0a635dc2a7060f : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/682b01c232974d0f9bb75081f4bbd558 : SUCCESS in 20m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/265e5582ce87443d83739790b7e6d592 : RETRY_LIMIT in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"3d9bc66029f695fe16af5788f173269ff0b27061","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-28 12:50:29.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 41m 49s","accounts_in_message":[],"_revision_number":39},{"id":"2790fee6e3979960d86b27a551223c337f11f3d5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-28 13:47:11.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/956058607439405fbc8d52b9773eef84 : SUCCESS in 42m 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/90be03d0eb6743e3aac052169467303e : RETRY_LIMIT in 15m 28s","accounts_in_message":[],"_revision_number":39},{"id":"eb509bfa2641f1f2403186f803535e8ead3d05f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-01-29 11:51:22.000000000","message":"Uploaded patch set 40: Patch Set 39 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":40},{"id":"45022c7c67783797adebe49db9043306ea48e42b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 11:52:41.000000000","message":"Patch Set 40:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/733/","accounts_in_message":[],"_revision_number":40},{"id":"9669e445edbaddc98a7bf7dede6267e0c979fae9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 11:52:43.000000000","message":"Patch Set 40:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/40/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/40/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":40},{"id":"4b23d5d543a6a7c56cf230b996926da36e9c5de2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-29 14:44:40.000000000","message":"Patch Set 40:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":40},{"id":"a4693737f594f798a5cfd02eb3571ff4a323e144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 15:06:01.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43e40224f26549b89fac87ec1d0fd97b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86269a1c11c04e7789a9865ab4c1933a : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72fe81b60bb94c57a1687aac8e31340e : SUCCESS in 12m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbc9735d330f4d809c3000bc8563306c : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26fa00e176a743ed9640452962f92c0a : SUCCESS in 14m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/c2dde9ff7b69447f8b800d3a9ba6d619 : SUCCESS in 1h 02m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/146730784d584a2dba34a0ca784cc9a5 : SUCCESS in 30m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3cca102a90646bbb37e6030d237c442 : SUCCESS in 1h 27m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44997fa71e0245209432f97e791b4992 : SUCCESS in 42m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51d4a0aa664f42ee975fbc0eff235d99 : SUCCESS in 15m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/25c01a88c990476ca26b290190de0712 : TIMED_OUT in 41m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4a1c3667a0d546c9a1ed3de20c66edeb : SUCCESS in 3m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5f1140e1c3e418ca60f3f9b69023b96 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e09b8f72319045e394a553889bca8288 : SUCCESS in 6m 19s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f7a29c6e31d643d0ab2f61a9d92c5b22 : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cfd829c9507b4f25bac90a7f4bfb4e2b : FAILURE in 35m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/92137826c9204c25b6c77d5da02f6f08 : SUCCESS in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0d968ffca0a64383b2a4be4f3b2ce082 : SUCCESS in 1h 58m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bb1c81f4abbb424ea36e382c85a3c976 : TIMED_OUT in 3h 06m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6aaf7e59cf0a4788937a5802902ac0b8 : SUCCESS in 1h 59m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/06139e11473545f7bc341716928bd65d : SUCCESS in 2h 08m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/afe587739ee94f0fa6711f5828636432 : SUCCESS in 36m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/175a667f713d42f39a2d184b23dd9abe : SUCCESS in 1h 14m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ef8f7d85292c4719aceb990326dbac5c : SUCCESS in 1h 38m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fd5c212279304b55802785c5f7979c8d : SUCCESS in 45m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bfb2588aa0684e8da6443c763c1de18d : SUCCESS in 1h 56m 56s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d47e7f98b42f4fdc9ac382cd4d8e71cc : SUCCESS in 1h 56m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/540e8b89a74e4cc89a201ce13b46b275 : SUCCESS in 1h 48m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/327b6230d191492e91891fd31bcd51cd : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":40},{"id":"971b92928def0f392930f6163bc4954276c81520","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-29 15:37:59.000000000","message":"Patch Set 40:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_01_2026_15_31_12\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":40},{"id":"423daf17066007f34e98bcbc5d8e60aeb0b60c31","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-29 17:31:35.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 38m 58s","accounts_in_message":[],"_revision_number":40},{"id":"3f8d908f6d8262b7dc9a65f02c5d15a3f521eed7","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-29 18:32:15.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a644132e1f554aec9b0487e89a87379b : FAILURE in 39m 05s","accounts_in_message":[],"_revision_number":40},{"id":"a2842e2f93181bfd556bad736d7c4fe9661f2282","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-29 21:30:04.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5d941de9ba9543eebd93f1201156f200 : SUCCESS in 47m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/73d59c0d216a4434bd5a7baff50c707e : RETRY_LIMIT in 15m 25s","accounts_in_message":[],"_revision_number":40},{"id":"7233abba1cc3d11a3423033ae4811565e9ecf305","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 23:10:12.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/358e8498101042ae81a582fa3f3a7437\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f7abc7487c6d4a07a199877af9d99f7a : SUCCESS in 20m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c006e91b66f14f309ed0c4e86b1a22c1 : SUCCESS in 19m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dbe4f3626cef4956af4b3253456cdf6e : RETRY_LIMIT in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"15e116fcec265629c4375765b8cbd62ae211dd6f","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-31 19:09:32.000000000","message":"Patch Set 40:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_424/index.html : SUCCESS in 88m","accounts_in_message":[],"_revision_number":40},{"id":"221b09e6f02b6a3cd31dccf82a3fb6ef27dce7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-05 06:38:56.000000000","message":"Uploaded patch set 41: Patch Set 40 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":41},{"id":"ceead0edd51982813e5ffe9b3339b2e2998ed8f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-02-05 06:52:25.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.","accounts_in_message":[],"_revision_number":42},{"id":"d164d2574899d06b5bfeef8c52cc06c0211c22ee","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-05 07:20:33.000000000","message":"Patch Set 42:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 975638,4","accounts_in_message":[],"_revision_number":42},{"id":"32890457a401b0ef6d745f3e2664d27fd2c303ad","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-05 07:20:41.000000000","message":"Patch Set 42:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 975638,4","accounts_in_message":[],"_revision_number":42},{"id":"cc45779ebc4d9ba1898f1af526a4832640b8e7d4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-05 07:20:52.000000000","message":"Patch Set 42:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":42},{"id":"36092170efd8ec9e66e64dcedb9432a8837b177d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 08:42:14.000000000","message":"Patch Set 41:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":41},{"id":"d8a2374e4cb4e30f495f174a24f17f90b81d0693","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 08:58:15.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a1f8436bb92449e980c8297cbee8b4c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f3d30eac496474f8e7ff55a0aeaaeb0 : SUCCESS in 20m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/503e3d8b0d184814a301c5cf832b4d0c : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c7dcdc83838d49aba89bba2f5aa243fb : RETRY_LIMIT in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"4cc23d427791b661a9bae92a8cad7b8dd65755d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 09:10:13.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ace3bd1abd74e70b440ba01eef89951\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75c256dc198342d7a799ce0c7b3ebcd2 : SUCCESS in 8m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/925084680b7647b8a818057d8a7d3c76 : SUCCESS in 12m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/759d96548d064508a23692d5b7d3b19e : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94b04ee2d02c4727bfd6393f0055cec7 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/f4207e27d617471b8ed5ef553f109c83 : SUCCESS in 1h 04m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc8937e831fd44da8dd39a7b158d087b : SUCCESS in 58m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b0fc1ec4af0e45e49a3dfa13a2fda607 : SUCCESS in 1h 20m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9efcd8631255403bae4d5c8495aaf79f : SUCCESS in 59m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51e9871aa79e4fc78a3635c936954650 : SUCCESS in 18m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/53e8f544c7af4c288ad981352712b3ee : TIMED_OUT in 41m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/77ac13083f744a15bc99e08299132a1e : SUCCESS in 3m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2a0cc48db52f4359be4154dd536975c0 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/77d8e846f8ab4988b901c4d6d70ce31f : SUCCESS in 5m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3e1ca56764894c5199675bc0668f81d3 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6ad4ef586eef4a27890077aa5632fa0a : FAILURE in 40m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/367cc84be2d1481b991bbbc53a0351f5 : SUCCESS in 8m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5e8549d8798c4aeb85fefb85814a6341 : SUCCESS in 1h 46m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/84fb2c92670e41298fa96b963d3d5ff9 : FAILURE in 1h 28m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dcf5d582b19a42c59ba3d3e09b7a55e6 : SUCCESS in 1h 56m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/715ba92291dc4169a64c90d4b974d238 : SUCCESS in 2h 08m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8a398dfb9a3f4a77991d16dc3a0ce1dd : SUCCESS in 27m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9923065491f042e186892edfebbe79e7 : SUCCESS in 1h 21m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/907cd137072b4b5990d72a779e2b0354 : SUCCESS in 1h 32m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6ec41fbe5e2440c9b2dfa6c20c459921 : SUCCESS in 26m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b6f5de46c1d946c28325b18e31ebe5f5 : SUCCESS in 1h 59m 51s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9a27b7ae78c84eb29aa14a13cac44d4c : SUCCESS in 2h 04m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8e1336bccef2412f88ddde63a60ca964 : SUCCESS in 1h 51m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e8ed4b1e5184e35aed21b433c635bbf : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":42},{"id":"00fe07a0a8ec83e5894054566f87a08ec5a2043c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-05 09:20:17.000000000","message":"Patch Set 42:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_02_2026_09_13_27\nLeave a comment with","accounts_in_message":[],"_revision_number":42},{"id":"d81527d2b705ea2c810c26684fb3b6d937f260a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-02-05 09:44:17.000000000","message":"Uploaded patch set 43: Patch Set 42 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":43},{"id":"a10611988610fbd237a79a24b170216f399cf7af","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-05 09:45:23.000000000","message":"Patch Set 42:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":42},{"id":"f0c9a7de83042f1d87f88a51815b4cfacab7918b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-05 09:56:20.000000000","message":"Patch Set 43:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":43},{"id":"a94cc696918e3587d8ecf078a55c6bcb2405145d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 11:38:32.000000000","message":"Patch Set 42:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":42},{"id":"4423be0ddcf8570f523e4a7a4e8ed49618b8dce4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 11:45:05.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f501377717da49abb62f06b7b5390399\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3626c2eab74d4063969d729d7559e638 : SUCCESS in 23m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ec044084f7784c489b199e3088b08710 : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f164ff03a4e74a1cb1e5e5701ba91a3a : RETRY_LIMIT in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"349adf232bd173c23b7cfaa360ce604eeed7b1a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 12:48:57.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c99c5c29ff4e4bbf9c3a00c74fdd9d21\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb1b3d09739a490986efdad866d8f203 : SUCCESS in 9m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb22626339ca405ca17748948ed7e0e3 : SUCCESS in 15m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25c55f28e4534fdc8f8de9c96ce69505 : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c4abc5018040cb9d218f5d47093cd5 : SUCCESS in 14m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/fe76715e964e4c47bbdc2ee6f8c266a1 : SUCCESS in 1h 03m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/102950cfecb04036a3c131f3f3a9bca8 : SUCCESS in 1h 03m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/29cf4b30764f45b1955aefc2e06de9a8 : SUCCESS in 1h 17m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e153fb7fe064a4f9218d96cb1737f06 : SUCCESS in 1h 08m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb2a94a8d4254e328d9eac7ca4b0e5fe : SUCCESS in 17m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0e335e6a27ea44feb9d4b790bf0571e9 : SUCCESS in 21m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9f7794df29f64a6185381e91aec73cd1 : SUCCESS in 6m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d19f9abe8ca44d2885b0a73dce3b933f : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6bff9c59bdab4b148251d1acfa3dcb80 : SUCCESS in 6m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0d9bfbc89b034174b99962816993935b : SUCCESS in 7m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/77508e0a876f410697c23fee16d563a4 : FAILURE in 37m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5bbbe215a02b4d189118d8ee1733a225 : SUCCESS in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6a40002745da4e62a7be83bc7aeab2d6 : SUCCESS in 1h 52m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d66d46b4f8514db0bee62804c7521390 : SUCCESS in 2h 34m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cff3818a10b04c5fade7f1f17704a5a5 : SUCCESS in 2h 02m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f9e7ed0281f84c9588971b2809f862bb : SUCCESS in 1h 56m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9e8f5609c8524471aaf5877b251e4592 : SUCCESS in 37m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2ec0bc6950174e56b3adf00aab614dcc : SUCCESS in 1h 14m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/71e6cadbbdb84b16ba904e683ac87f8e : SUCCESS in 1h 48m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/629942b5745c49e99e48cd46db733d14 : SUCCESS in 47m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c0a8855e27bf48ada2446ed1f59ba84e : SUCCESS in 1h 56m 16s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/dd0c6033cc4e4df8931ca92263e1b6c9 : SUCCESS in 2h 14m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/17b0f03d95aa4842aa6787b339f20541 : SUCCESS in 1h 49m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/842676e69c694eb8a4113a0b609ed2f8 : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":43},{"id":"7e602600ceaec96660074b368fc245d8fc8321d6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-05 13:24:51.000000000","message":"Patch Set 43:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_02_2026_13_17_40\nLeave a comment with","accounts_in_message":[],"_revision_number":43},{"id":"aa4c3016df3d615af6e7e4ec6b844007b430f48f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 14:55:38.000000000","message":"Patch Set 43:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":43},{"id":"3dcf49110bc7d0c0a141bebeccf85d3b8f85147e","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-05 16:05:47.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/d3ba751ea94a4b1495a7f37a3660d76d : FAILURE in 39m 35s","accounts_in_message":[],"_revision_number":43},{"id":"899ea25a0e567cb4eba07ccfefd0734e9aac41e7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-05 17:35:45.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 40m 28s","accounts_in_message":[],"_revision_number":43},{"id":"00003b57b40b3673a2fd8f2a6051d6ca327763dd","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-06 15:05:16.000000000","message":"Patch Set 43:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 938771,101","accounts_in_message":[],"_revision_number":43},{"id":"50eefbcc6ac4f9212c025fb161805a20682d2d23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-02-13 09:22:24.000000000","message":"Uploaded patch set 44: Patch Set 43 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"9726c30216a05074deda25602ad2980e1250d81b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 09:59:53.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac2358ad93148fc9e7f23c96b90dd58\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67dd79f5fc4544888bae5e3ce3fae045 : FAILURE in 18m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba0279f032fc4c06be523faafc565544 : FAILURE in 10m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/efd66dec40da46cfa3f8f5d2e1282525 : RETRY_LIMIT in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"7266b6955bd8d9ed078d9f09ab46b514f862c19b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-13 10:36:37.000000000","message":"Patch Set 44:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":44},{"id":"bc79e3c973c25bc6f8ed241cfc67b2d482826fb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 12:19:08.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb98dcb709f543b9ad679b2762ec86d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2b10c45ec8c4829a2469754033ed3f0 : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97be6e8ce76944dcbba0a91d238e0924 : FAILURE in 7m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5ef67b1ec4d477abd3f97b0cd394029 : FAILURE in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/175fd8e07e874e09b0b24496f166c71d : FAILURE in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/bf0deebee6034c51995d48f80b396de2 : SUCCESS in 1h 03m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/31fea07790774157adb869c1e2f68454 : SUCCESS in 48m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3b4991255bc44db0ac2331b7efaa434d : SUCCESS in 1h 33m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c400b6c72d17484fa91197ac20bfd5f1 : SUCCESS in 1h 03m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b6cf9db95904b2ba8cc61eaeb461115 : SUCCESS in 10m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7457e6b765484fd4818d2fe030511443 : FAILURE in 6m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a7a8e3ff5c114a3fb643603212eb0094 : SUCCESS in 3m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b6e97d2af8e14433a4e276262f09cc0b : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9ab93cb3b204040830c411efc4cdb59 : SUCCESS in 6m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/46587489424f454f9a0dcdd61c26fafc : SUCCESS in 4m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c370ea69090245a993f1cf7d6722376a : FAILURE in 39m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff0e1bd89c374b4eb7fb1b0836bbf245 : SUCCESS in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e9d64554cc2f470ab5486010b8f71723 : SUCCESS in 1h 30m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bc0d858619fc43bb913251aa7dce12a1 : FAILURE in 2h 37m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/27a7715cddbc4e3ab1742e401c05cb66 : SUCCESS in 2h 01m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4ed4e44481fc44228ae77f631fb838b3 : SUCCESS in 1h 48m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/59d6097371e1413281990daaf79e9add : SUCCESS in 37m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ebd7e4cebbfd47fca97bae64d31dec07 : SUCCESS in 1h 14m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6c261b35bf884328979ca28d1524125e : SUCCESS in 1h 48m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/58033c5722e14bb69366a4ada433e95f : SUCCESS in 47m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ca63c96eec4f4b7483a58c8e2579f145 : SUCCESS in 1h 02m 38s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/eefcf1b6966d427e876d6a75939268ef : SUCCESS in 1h 18m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/614a8a8496ed4cc59a37ed13a0558207 : SUCCESS in 1h 41m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6efa52ad89bc47c297d09a0732970b15 : SUCCESS in 1h 07m 23s","accounts_in_message":[],"_revision_number":44},{"id":"ab0683e259ae3c3fcadc10761dd4f1fa75f31ddf","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 13:28:51.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/bd144f707852494898024f12e982a8e4 : FAILURE in 24m 03s","accounts_in_message":[],"_revision_number":44},{"id":"c3e02d968312a1cc9748ed6e5a1c2bc55c31bfb6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-13 15:31:06.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 36m 47s","accounts_in_message":[],"_revision_number":44},{"id":"c652d5ab83551fca627d39d1edacf8380052ea9b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-13 15:43:08.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a13076dc485949339b9998914c38abe8 : FAILURE in 39m 47s","accounts_in_message":[],"_revision_number":44},{"id":"3bcb4e7e5d322631470d00608b4b25a065973dfb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-14 12:37:41.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ecfd47f5619a45e9a1953218d747af6a : FAILURE in 1h 03m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/201599c7c52b4ccdbc873c00739bfe44 : FAILURE in 57m 45s","accounts_in_message":[],"_revision_number":44},{"id":"03463cd8f7965d0526c0b63a11ed59422633dbfe","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-14 16:40:08.000000000","message":"Patch Set 44:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_644/index.html : SUCCESS in 99m","accounts_in_message":[],"_revision_number":44},{"id":"0fd8fdbc09bb9c015e2ec5412416d23397d7f21d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-03-31 06:18:14.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":45},{"id":"5833d5eb62fba5fb7e8b5cdcc85e6a42698ac8f3","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-31 06:20:05.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/e6e09e044fb44845b910ce7a435fb64c : NODE_FAILURE Node request 200-0000692609 failed in 0s","accounts_in_message":[],"_revision_number":45},{"id":"46f0348c3a35e46fbfcbb4ec78551ee4706757fa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-31 06:28:05.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/64f5f6574c794f98ace19ccfeaae0c1e : RETRY_LIMIT in 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7690e429b23e4095aaa0d3fe05760a79 : RETRY_LIMIT in 39s","accounts_in_message":[],"_revision_number":45},{"id":"cb54530db6ecb8012d511303b62aada80b892138","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 06:52:11.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b878459d214f47aa8aa6b48888a1faf0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89a204cb9dc54c0084099149c42ff367 : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/895d5c8cea7a4732800e444e6141c035 : SUCCESS in 16m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7c13ecf3c22b48cb8b77f1920ed0f89b : SUCCESS in 22m 00s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"c7bbd6ecfe59860d90857eccb45b07daae9fccc5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-31 06:55:51.000000000","message":"Patch Set 45:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":45},{"id":"163dd32201ee2ba163343b09f41047e13ad84f53","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-31 09:06:02.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- hpe-fc-driver build/d0e6e6f8f3ba40e79d373b558de2b865 : FAILURE in 34m 19s","accounts_in_message":[],"_revision_number":45},{"id":"0a535bee7b52d5df6304e1af59cbf6003c299c42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 09:34:20.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a62aa53ce7041b08801885d2b352fcf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b574c3a7701c4d368a7bd6e43c59553e : FAILURE in 8m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/381a55a07fe742ee864c78f3d5ab310f : SUCCESS in 10m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1688ea3c6993492cad0e0954dd0b1431 : SUCCESS in 10m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c29015e970ee4423bd3a09bbc697fd7d : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d83c0d173b4431b754809881c06c0f : SUCCESS in 10m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/d6bba2ae9ce646f0bf35c293449b4a29 : SUCCESS in 1h 11m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5347470f140a42dd8f2439e4d0f83eee : SUCCESS in 46m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/012cedeacdbc4babb71a53f3b5426abe : SUCCESS in 1h 39m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee230f3f397a4c7b81ba04e5fec0ccb3 : SUCCESS in 38m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d55fca497c1418aad329ea580a3470c : SUCCESS in 15m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/375d90b38a6948fdab715493b1a7e2c2 : SUCCESS in 26m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6d8a33a587d4489692dc9c4e7da4ea49 : SUCCESS in 4m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/325dbff40b324dcc8cf7a413d0e7cfbe : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5de4dd83c4d44c028c637ce91ae825f4 : SUCCESS in 5m 40s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/867a5ff81c404968920ff961bcc871a1 : SUCCESS in 6m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/60db2269b807487298f15a88e515d10e : FAILURE in 19m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/68b61f0549d642bea008b59260b9e8e7 : FAILURE in 4m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/10270509a49447999b46d2d3e23c4514 : SUCCESS in 2h 04m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6477d50afd234b0dbd19c6901218fa88 : TIMED_OUT in 3h 06m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6a6e0cfd08b142088498ff94defda29f : SUCCESS in 1h 47m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3859867f8d0742999b5288e50cb982e0 : SUCCESS in 2h 07m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2bb038041f35405dbe02e7c7454c7034 : SUCCESS in 36m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/969ac33d89b144458c3d91aa9e3ea174 : SUCCESS in 1h 04m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c30cf6ff554849efbf43c615cf02887e : SUCCESS in 1h 39m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a84dd43f92b54ac682a6ae0bac44d97c : SUCCESS in 48m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/de4ae45f718545758b179483b36fcaec : SUCCESS in 1h 57m 19s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/1ef2ce88e0ea476ca6d80655d9e6ac2b : TIMED_OUT in 3h 09m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a73f088a91b24d38b7d85b2da793850b : SUCCESS in 1h 32m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da6b0e9bf06645cc8ce0451766100cef : SUCCESS in 49m 50s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":45},{"id":"3701b60f339ee9aa26c66c56d411ebda5bb90c00","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-31 09:40:19.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 39m 44s","accounts_in_message":[],"_revision_number":45},{"id":"94ee59aa95079ec2bf604bbe08ed37555bfb3f62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"real_author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-03-31 09:40:48.000000000","message":"Patch Set 46: Patch Set 45 was rebased on behalf of \u003cGERRIT_ACCOUNT_36800\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"}],"_revision_number":46},{"id":"987ae6cee49ec541e852e8f40c2247509834743d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-31 09:43:40.000000000","message":"Patch Set 46:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/8454daf7196c4195b07d284549912f8b : NODE_FAILURE Node request 200-0000693075 failed in 0s","accounts_in_message":[],"_revision_number":46},{"id":"414fc1fff0168b16785febf0bb50a927fb5bb182","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:05:36.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c75fa18ae35498fadf35c98429de5cf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2dda27938cea496cb571c02779990266 : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c422bf488f7b453184e8af52accbf3c4 : SUCCESS in 13m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/829b08c0889f488393d0c503e823b364 : SUCCESS in 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"cb9265a2e4afafa5ea59ebf2bf110e83e47f9f22","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-31 10:08:06.000000000","message":"Patch Set 46:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ba579c56c42d4bce87940999d4a127b5 : RETRY_LIMIT in 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3db0a49d978946c8847cc4c0799e3f87 : RETRY_LIMIT in 39s","accounts_in_message":[],"_revision_number":46},{"id":"ef7983a8ccae1b2b13404fc835c7bb0eefae9a21","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-31 10:40:03.000000000","message":"Patch Set 46:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":46},{"id":"0379931d84af5e5518d1f34a5731e11b7ef1c677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 12:23:41.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f38136d40c6d416aa1fa0db6cbc93b46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5f83c06926d49c99c81716b23be1bce : FAILURE in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c60cb8c5dee641eaa25647c135a40974 : SUCCESS in 11m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb14f5fdb23a428f9a359dde9bba0d5a : SUCCESS in 8m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e0b8c6395ef74020aafcd95dcfb42166 : SUCCESS in 15m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5978585ef00445cbc82b7eee52adadf : SUCCESS in 14m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/4c5451ae4b8a4b7c9276ee0aa2a8a342 : SUCCESS in 1h 01m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0bafe475eb740659e18a76474fd7667 : SUCCESS in 1h 01m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f80c7b8038bb4648860848951fbad729 : SUCCESS in 1h 23m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51f44a195a724e2e90e1af92a639538e : SUCCESS in 40m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6774f53603774c4caad9bf3844df573b : SUCCESS in 25m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ccae7ab767c54a2f8eaba71decc34b99 : SUCCESS in 22m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2da27ab791b540e8a9534806e213c3db : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/983d09cf64b24e43893f9fc53dd18798 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/950aaaf5a0374ff68f8a0feb925d4654 : SUCCESS in 6m 19s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d84aabd79fc64b61878de98869e67b87 : SUCCESS in 6m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/46d60812619f423380232ab83da0172b : FAILURE in 31m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7d080baff1e84a91863191b81ad97b92 : FAILURE in 5m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2dbef42978b443ddb3168e4949f8b0bb : SUCCESS in 1h 45m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b51716b211914989a275ac9272ba6ec4 : FAILURE in 1h 43m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6857aba38dd74234a01c401a34df195d : SUCCESS in 1h 54m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a475808667d74b8fa0add1b43ec33df5 : SUCCESS in 2h 15m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2b440f45a493493b8329a4f23cbd3519 : SUCCESS in 19m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a8cdb58873494390bfc7ce27b7709d1d : SUCCESS in 1h 01m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/48ef496a931e4db49c74aba70ce86582 : SUCCESS in 1h 46m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ee06f9ce7e1c4446a2b5f4b022b82585 : SUCCESS in 43m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b9d7ec32866d46bebd62be3663fcbf55 : SUCCESS in 2h 01m 08s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ad607bdf58414ba2b3b1917d54f0ef07 : SUCCESS in 2h 35m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1fa72d9b98e642849e1efb26fd8b57cf : SUCCESS in 1h 29m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7718034067574fc8a98284a2d84b5cff : SUCCESS in 44m 32s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":46},{"id":"cebb43bf2c40633f4eef34ba588844b9d08241d9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-31 12:52:55.000000000","message":"Patch Set 46:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-46/FC/ : SUCCESS in 2h 45m 27s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-949483-46/iSCSI/ : SUCCESS in 2h 48m 27s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":46},{"id":"31893a086794ea60a743a5c57e1039de5c229878","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-31 14:01:08.000000000","message":"Patch Set 46:\n\nBuild failed.\n\n- hpe-fc-driver build/52a36f41d7da4b7c89da94dd6ffdca38 : FAILURE in 34m 08s","accounts_in_message":[],"_revision_number":46},{"id":"18e59ef2133259ff0a71b770de6fefffe3626ce9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-31 15:16:15.000000000","message":"Patch Set 46:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 43m 11s","accounts_in_message":[],"_revision_number":46},{"id":"909d6a04f32395c9ab0193d4ae2dcb25a106e120","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-31 15:33:58.000000000","message":"Patch Set 46:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/9a35e5666b3d4000ae5e496b04001d60 : FAILURE in 40m 04s","accounts_in_message":[],"_revision_number":46},{"id":"618daf11074b1effa8b5c7b0adb460d06de1fa07","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-31 23:51:26.000000000","message":"Patch Set 45:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3640/.","accounts_in_message":[],"_revision_number":45},{"id":"fbea421a7e6960e7dcc290e747d322ae40438912","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-01 11:29:16.000000000","message":"Patch Set 46:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3650/.","accounts_in_message":[],"_revision_number":46},{"id":"c470f1846284a7337656f5ced67642284cd5c82c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-03 18:04:36.000000000","message":"Patch Set 45:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3467/.","accounts_in_message":[],"_revision_number":45},{"id":"d309b8f061331f42765fddd183d7147aef966b2d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-04 04:09:28.000000000","message":"Patch Set 46:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3477/.","accounts_in_message":[],"_revision_number":46},{"id":"020d3a6f385e52bf8f7a8849f31f72261c086dbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-04-09 10:19:31.000000000","message":"Uploaded patch set 47: Patch Set 46 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":47},{"id":"f574c04c3c93329a551b09beae711397f24489b4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-09 10:58:08.000000000","message":"Patch Set 47:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ce4cd3a58c9146d98eca98d5db3f593d : RETRY_LIMIT in 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4e72983541b94b3ba77abc245cdec3b3 : RETRY_LIMIT in 40s","accounts_in_message":[],"_revision_number":47},{"id":"fff76a7d4be49386070c8d7912a2bc67223f3aab","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-09 11:02:54.000000000","message":"Patch Set 47:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":47},{"id":"b0c44ab87e81de6fb7cd22bc2c03b9d80cf9eef8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:53:52.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2823821703ad4815b38ef9e53e0b0a82\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/17f84f85aea140ef8f87efe03bf0b131 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6a1179f8594f45dcaf27834ee76c0888 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7538e60e5cb6493095567e10a2cc8b60 : SUCCESS in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"cef79e90823926a75c6a03167bc1cd5506938d85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 13:43:43.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/465ba6a1bd37496bb272968695e75787\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/330a7fce36e0438ba2b22a0405716dd6 : FAILURE in 7m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c5e24f5eb1d401983851d00b5d00609 : SUCCESS in 12m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9b113df04ee4b858971282d3926f916 : SUCCESS in 10m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2466fbebb7c944d780e00a942ed6522a : SUCCESS in 13m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ec94221c734003a23b7fd30dbd9e5d : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/bef6964ad3d142b2bcc8902fb758998e : SUCCESS in 1h 01m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/69e58f76e4dd40518c6e830a3b5d1346 : SUCCESS in 1h 00m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3eda669dd83d4549a0d870070d6cb9b2 : SUCCESS in 1h 35m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb6b6bc0c37a41dabd79ba189e7a01a6 : SUCCESS in 56m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e33613edccd24bbab6a2f29ddf710226 : SUCCESS in 21m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e6dc2f84e1c47d6bcc31554b9b15da4 : FAILURE in 26m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/780a7cf5de844158bd64eda23f418c16 : SUCCESS in 5m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/38c459b85cec439ca0ed0e365992946f : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8498346c6811442996672e646f70e69e : SUCCESS in 7m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0df0609cbc294ee68985262781b4d198 : SUCCESS in 7m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cd7182ad202447da8baa7d5e95831d68 : FAILURE in 41m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/edecbf809ce646778a11503bc3b26f5e : FAILURE in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/64d39058a5b74b93becfc57d79a458ef : SUCCESS in 1h 49m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2b00828b51b04cedae7151b26f534ce1 : TIMED_OUT in 3h 13m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc0941d56ed94661b1e3fe27c42f65ff : SUCCESS in 1h 33m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/40615b83094840c2a87e0b7ad4464fba : SUCCESS in 2h 52m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5efebc8875ee4e7797591bc968763353 : SUCCESS in 38m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/02de6092b0ab4f30a1061509b8e02e99 : SUCCESS in 1h 16m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/282ae95273514bc6ac7da768db15338b : SUCCESS in 1h 43m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5a1ce296bc6a4d5f901a94c63e1df3a3 : SUCCESS in 31m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a38d53a8441c4cf29c5410ce3b0b2410 : SUCCESS in 1h 30m 36s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/14a4b88d2f634be9898e199901242fd5 : FAILURE in 2h 08m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/20640c4344ed4435b66704dd520fd3b0 : SUCCESS in 1h 35m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2126754215d34218a58cfb4736f94b82 : SUCCESS in 40m 18s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":47},{"id":"441b5b6651850020687015e2b150132b07357f2a","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-09 15:05:58.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- hpe-fc-driver build/7179ae07bf584da39f890c416951331b : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":47},{"id":"8078a62f03169ca38e3ed5e6b38b67093e57e3b3","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-09 15:38:54.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/cab603d1fae6403c91286afca1dbe58e : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":47},{"id":"e6622a4985c6716417cec1daabd3be8b75fd5e53","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-09 15:49:05.000000000","message":"Patch Set 47:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 37m 51s","accounts_in_message":[],"_revision_number":47},{"id":"d71ebd75920bb258230bdec5a6151f2f827ff5b2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-09 17:27:16.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/937095bc372f4393bfa6d2483be6868a : SUCCESS in 47m 09s","accounts_in_message":[],"_revision_number":47},{"id":"e13cbec3a0ed86a23a042e8af3a300f1d2a3bdf8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-09 17:57:11.000000000","message":"Patch Set 47:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3561/.","accounts_in_message":[],"_revision_number":47},{"id":"7e7cb5fb6634e63472e6477eddbf3f76b01d583f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-09 19:47:30.000000000","message":"Patch Set 47:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3737/.","accounts_in_message":[],"_revision_number":47},{"id":"9e38ce6d10bb22396a0735686238b1be3363eb25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-04-13 07:27:00.000000000","message":"Uploaded patch set 48: New patch set was added with same tree, parent tree, and commit message as Patch Set 47.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":48},{"id":"782e8cc241250ef1c578cb7fbab613b132d6ab77","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-13 08:09:10.000000000","message":"Patch Set 48:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":48},{"id":"e672855e7c4722c1e0d04ddd0add671f2b32d276","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-13 08:33:47.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/25197c73a35c4e42840f43a721771589 : RETRY_LIMIT in 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/92f49cc38f814aae8097e58b22d6a5f9 : RETRY_LIMIT in 39s","accounts_in_message":[],"_revision_number":48},{"id":"b26d90fb8cefb23f82b1c9d42535ef129560f7bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 09:07:49.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9823b315c810459a845f91ec00b48ed1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/218baa099a394a7a9cd2e73f4ab13135 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1740ea2aea86434eabc282b860c20273 : SUCCESS in 14m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/75c1ea8f3b104153860ca08080efe3a1 : SUCCESS in 16m 33s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"742271aec8db680bdf8ebc57965635171a90ad69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 10:12:01.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ef461b1641b4b1a8b4eca49099e09b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bea3ba40b2ab4863a4ccd49bfb55df6a : FAILURE in 6m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5ac2857c777437b831bcef3ca5ba059 : SUCCESS in 10m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abbb14918484423782e720e99f6301a0 : SUCCESS in 7m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3602a83b136b49d487e1262bc0d04233 : SUCCESS in 12m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22237b3c4d3346fd93e234d0505e7cb9 : SUCCESS in 15m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/45c0b88b3a444032b71b18c551d2c47b : SUCCESS in 1h 15m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/27cb58cdfcbe442a9b309b6c26d09701 : SUCCESS in 1h 05m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b28a5688cc84963970ec079733d942e : SUCCESS in 1h 43m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c5f7c1b37314331931b291ca79c4891 : SUCCESS in 1h 09m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e534f5112ae4720b146b2780e4ffb9a : SUCCESS in 18m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8bfe5ad28c6248618fd291dcd79125d1 : TIMED_OUT in 41m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e80058fc3b434322bce72f787173605b : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1e555bc6743c4cf8a53e30b940f00ddb : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95f2dd0635fe45f9a563e04053851039 : SUCCESS in 6m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a59252e42144472cbefcea75fa163903 : SUCCESS in 7m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9169b86093545918b7279640ac3d9ec : FAILURE in 25m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ecba97f602f4c90b5f88b360f4e8414 : FAILURE in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7a76a35a09874690932fe124c5686cb2 : SUCCESS in 2h 14m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48652193e91e416b9a2d973b996149d3 : FAILURE in 2h 19m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/499e1b21f67e4544adb325a38f28f41a : SUCCESS in 2h 09m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4e9cafd7ecc54516a1318ff6ea291750 : SUCCESS in 2h 08m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/34e98ed7e3d749d385cf626840f31b67 : SUCCESS in 28m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c9ce10aa3d404a6492645da0b2793b1a : SUCCESS in 1h 22m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/38e26a19601d4ed886024cec8c75acfe : SUCCESS in 1h 38m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d2f4db3ddabc4c859f4e93c402ad4459 : SUCCESS in 46m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/71ad0648ee8847bbb8706a1a073e88d9 : SUCCESS in 1h 08m 42s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/5fef1920ba6848399a3da0d8679d6e65 : SUCCESS in 1h 14m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d835fdf9478c4c8ca54226d3a515e2bf : SUCCESS in 2h 00m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09cdd1dad97346388140b0001188985c : SUCCESS in 1h 00m 50s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":48},{"id":"6e4bd0d3ad212f10d923da3a198f52abb3c59706","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-13 10:21:07.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- hpe-fc-driver build/9f4f521fa82c4ac69185277eaa3a1a64 : FAILURE in 6m 11s","accounts_in_message":[],"_revision_number":48},{"id":"f23189d74defd0e8aeab2ffbd54f901ff22b9990","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-13 10:24:09.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/b5b6cd431f224f6d942e5f79040f5290 : FAILURE in 6m 15s","accounts_in_message":[],"_revision_number":48},{"id":"d0ea16b0ca26e36738b3ce7c68ebc3f8f5656fd4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-13 10:51:18.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 13m 54s","accounts_in_message":[],"_revision_number":48},{"id":"126eea764e3519a0e2df6570165ed1655cd8a527","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-13 11:16:02.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/06799786e6824816ae12bd8a156f51cd : FAILURE in 15m 10s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/949483/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/949483/etc/tempest_log.log","accounts_in_message":[],"_revision_number":48},{"id":"e475bb8292af15152fc83233ba75bd9624b4c01a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-13 15:12:54.000000000","message":"Patch Set 48:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3766/.","accounts_in_message":[],"_revision_number":48},{"id":"47d2ba2d0f8577bce3aee1a4111b18871a7676ac","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-13 15:26:02.000000000","message":"Patch Set 48:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3590/.","accounts_in_message":[],"_revision_number":48},{"id":"458d081cb0ba5f36812532b50910ba99e3dfb7bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-04-14 11:04:51.000000000","message":"Uploaded patch set 49: New patch set was added with same tree, parent tree, and commit message as Patch Set 48.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":49},{"id":"02476de87cada63b8f9a95ab1334358ece0533b5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-14 11:49:41.000000000","message":"Patch Set 49:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":49},{"id":"4bacef6f6e6c994548a515d5f08e95d5aa9003ef","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-14 12:05:08.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d732ae5fc630400084458b6238edea62 : RETRY_LIMIT in 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0900e87571e04d14b76bd7ff4a16cf60 : RETRY_LIMIT in 39s","accounts_in_message":[],"_revision_number":49},{"id":"4028d14a05dc6ce2e6bab6bc84972ffc5a2fa89b","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-14 14:03:16.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/06e1cf3ccfdf4f9d9fb0feb2eda4ce10 : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":49},{"id":"c6ef1e49d21dcc9aa7568d25a2805ac744c2efb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 14:04:30.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bf823c281ef4b449a21df7540b7abdd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8f380c11265475f8b99b2350e0afe25 : FAILURE in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf6f8b94634f40468f87771d4f2e05fa : SUCCESS in 13m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/814bcba39e1a44d9b9d7bf8c49fe0f0d : FAILURE in 12m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/20d6ff446c8247a98ec968d5d3632fa3 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc863b049244aff813b7f52e8acb12d : SUCCESS in 13m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/f035f20381f74db7a8df1849408dfc07 : SUCCESS in 1h 11m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4f5d7a940f5440f5a0ea945d2253d0e0 : SUCCESS in 37m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88fe907a5b1844a0800a7c529f2e9900 : SUCCESS in 1h 43m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27cd56d3c2f146d2aa02130d819e65bd : SUCCESS in 1h 16m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf426bd60910412b8a5fde68ead26f91 : SUCCESS in 16m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8d78aa7500f640f09ef77009be328601 : FAILURE in 26m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/66db0236f57a48268a4023a4f6ff5033 : SUCCESS in 4m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/72a9878427e14071b494a0334393b671 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/187ff41fa3e341fbb109e7cf3f24a1f8 : SUCCESS in 7m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8adbb33b6dee4cd3bd3587f3e28ed8a0 : SUCCESS in 7m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a00a5ea7b344470b9f0b43af5a52e682 : FAILURE in 54m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/90773d6835a0474b9db92fba2f488ce6 : FAILURE in 6m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ca1279bb459648ceac0eed7c4504a678 : SUCCESS in 1h 36m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/894385c8c2484009bcfe7b11d7e86899 : FAILURE in 2h 45m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f3e82b8cb0b84e94a8142c9a93a1502e : SUCCESS in 1h 56m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/05928977924d4c6aa8b202038cdff5e8 : SUCCESS in 2h 24m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d33a680d81aa4b3688ba296e40b88cdd : SUCCESS in 34m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6b2f678867cb4e8081a2c223c192fcfd : SUCCESS in 1h 19m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0bf474dd46184458bfef8f8f9cc7ba1b : SUCCESS in 1h 32m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a8d75a8f24024ec59f56e631c6cd8093 : SUCCESS in 28m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c2cc4aea26a44be5af37cc6500b8cf46 : SUCCESS in 1h 54m 05s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/7d6528fed71a4653b1ae3c868867f6ce : SUCCESS in 2h 11m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e8c2aba25bcb4a7ab8942a43e1b82479 : SUCCESS in 1h 39m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15e3c0fbee344e168a936743ae3f8a2e : SUCCESS in 35m 43s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":49},{"id":"56504e1594facf6e859ba3121dae8b56218c7a34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 14:12:31.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/981d62b9e0f74af9ab6a28378eb16272\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/607cc80e5de4427a9d9eae274780e9cb : SUCCESS in 16m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9cb0f4ba843d4d3d800eb915a1aff405 : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/065a2f99b58b45409beb32e962abec53 : SUCCESS in 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"c53f3dbfb9877b8dd7a98d6df2260b73be6827d3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-14 14:21:29.000000000","message":"Patch Set 49:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3620/.","accounts_in_message":[],"_revision_number":49},{"id":"3b4bc3005688d860945aefb22dd81342301f727d","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-14 14:42:41.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- hpe-fc-driver build/29178bd696a442d48974d4d0357e7ae7 : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":49},{"id":"8583f53babbc44b6884164214a059a4258a1a094","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-14 15:16:25.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949483 : FAILURE in 39m 52s","accounts_in_message":[],"_revision_number":49},{"id":"e872c39d85f6d1ebb5d67df5ca0b0b40d22713e9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-14 16:54:54.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/0a8470d8fce74184a4b76987f005f109 : SUCCESS in 46m 11s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/949483/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/949483/etc/tempest_log.log","accounts_in_message":[],"_revision_number":49},{"id":"c9258daa473a8ad1382384c78b942b0885bf076a","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-14 18:02:04.000000000","message":"Patch Set 49: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":49},{"id":"30db547d4cfdb57610daa51f41a51e0f64fd2fbf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-15 07:57:29.000000000","message":"Patch Set 49:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3796/.","accounts_in_message":[],"_revision_number":49},{"id":"0606485e707248005ca4ab9b44f2b5a21661114b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-04-28 17:53:21.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":50},{"id":"95f9008cf358357b8a99274e7e1d737369ecc771","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-28 18:45:46.000000000","message":"Patch Set 50:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":50},{"id":"c30d35f86d212b73ce8bb2a72c9ee8c1814de000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 20:26:27.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/890afec5ea7a45adb8a16609ced16202\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdacb86357594f5b86be442a204a1f6b : FAILURE in 7m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4289fb047d20405dba72eae4c6262e0c : SUCCESS in 11m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/efd4b45550cd447fb1b207ee72575ad3 : SUCCESS in 10m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1183253111394431945bbbd644104c45 : FAILURE in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6f3d2b487f4eedb2d206d8ff7d365e : SUCCESS in 14m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/a7ba645e48de44eb88de6fa8e44672b0 : SUCCESS in 1h 03m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/65e7f8712df945c7bdd56b9c1d3517d4 : SUCCESS in 43m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e762e3a1d3004aca83da214eddb87f49 : SUCCESS in 1h 49m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eac3dc9656e6420faebd369fa1da71e5 : SUCCESS in 1h 05m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38b54546e27d49ff80cbff3ae7b3e86a : SUCCESS in 16m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7f80796cfeae4a1f8eadbba5c24887e4 : TIMED_OUT in 41m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6080f4c76c694e39bb5d3732d0b988de : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/055761329faf4c2ab7fe2b642d1f1913 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9845db275ef246e5b780e96b491012ee : SUCCESS in 7m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f115d3c8716741d2963eb91295f74c34 : SUCCESS in 7m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8ce95086eb644be3b262970a77778eba : SUCCESS in 39m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5d5f5c3f36c94de089613b8dea47a71e : FAILURE in 4m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a9afbbbd30624a7a860b4b2c083d1757 : SUCCESS in 1h 45m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bb08c6fe05a54fc5a60041582c7fab7f : FAILURE in 1h 56m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6eda9392e6b5467b92e215d53224fa11 : SUCCESS in 1h 57m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7d92da4e72a24377aeec717f136f535b : SUCCESS in 2h 18m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/24cccc11735648c5978f808074e6a225 : SUCCESS in 38m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7b3b5de840e448df8ac0c42441530633 : SUCCESS in 50m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/93c20c33b6624d41b7d51d2e2e7cb556 : SUCCESS in 2h 11m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/595de1ba599d434089649f2d059a9f84 : SUCCESS in 31m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7168d3b252ae482a8802780716f4ec7a : FAILURE in 2h 05m 01s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b029ccfcd3c41fc849534491e4a5921 : SUCCESS in 1h 52m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f87ff87a2fb4eabb54e43db9639054e : SUCCESS in 1h 01m 26s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":50},{"id":"17c0d73c1e74d8b8e787fb9ea7f7e3512828b537","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-28 20:58:59.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/dfa007ea0b9b4affbb1debd9c9bdc4fd : FAILURE in 22m 30s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/949483/etc/tempest_log.log","accounts_in_message":[],"_revision_number":50},{"id":"1ec58576c8efa6eb51df54c1b3f0efb1ceab07ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 21:13:18.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a024f19c2152415aa4cbaab09399ba68\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d3f6c3b815d44fc480963c47b4c1b10d : SUCCESS in 19m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0056dd96dd0941c8b865ad794c291988 : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/177e5d976c7e45c7b2ed9deb21be8809 : FAILURE in 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"6fed5a3473c40cd39cef18b57140b4d4e4f5f554","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-04-28 21:42:01.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- hpe-fc-driver build/f15808d43cc2424ab456b8fd9d029091 : FAILURE in 39m 17s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/949483/etc/tempest_log.log","accounts_in_message":[],"_revision_number":50},{"id":"c6fa4678ee3c70274e03a245a93da73affe95167","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-28 21:52:49.000000000","message":"Patch Set 50:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/949483 : FAILURE","accounts_in_message":[],"_revision_number":50},{"id":"90efcdd46ca7637947623aa08c35705c10f8a09c","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-28 22:03:43.000000000","message":"Patch Set 50:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/949483 : FAILURE","accounts_in_message":[],"_revision_number":50},{"id":"22f6dc2ca841b934aac30c18c5d825051f21b5b5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-28 23:58:43.000000000","message":"Patch Set 50:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3991/.","accounts_in_message":[],"_revision_number":50},{"id":"f96c1af468745c83ddbd621f534979f8a828ce43","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-29 00:14:19.000000000","message":"Patch Set 50:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3815/.","accounts_in_message":[],"_revision_number":50},{"id":"718e2bf22fb4af36d3d7b57ae17996b0bdcd9cb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-04-30 15:08:44.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":51},{"id":"92f4f1684025908790ac4fa8eb2797a060d72f67","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-30 16:18:16.000000000","message":"Patch Set 51:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":51},{"id":"2c8391a791cafc89c3212add286e5b6b2c16804a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:01:51.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63d51df05bee4aff871f4068f68eb8eb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78f31e7b5cde4b24ab6c7cd047c247a2 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb9fd3ed6a6a4e95abf39df416a64c99 : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ef1aa9c74ead4c56adfe846953611903 : SUCCESS in 22m 38s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"5cd50c9e08f15488b97708afabcf8d31087e821b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 18:11:05.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bc37343fd6b42cca30f16c1dcf25309\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b060e29fb2c479fa14572225548ac5a : FAILURE in 8m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e2b357beceb4b5c94c3805a39756c41 : SUCCESS in 12m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc4754d05adb49ab860b7a212f1403bf : SUCCESS in 11m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/531aceee71384fc98b7e6c095aa7e734 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49bf2bafc9e044ada2aa18d86a17dcdc : SUCCESS in 14m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/4dc50e8bb08640498769b2e639d1439b : SUCCESS in 1h 04m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/65d885a7704a4e0c8c133840ae5c2cd3 : SUCCESS in 32m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/619a836d09e7476eb65ca3c8d383aefe : SUCCESS in 1h 30m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a29d11865c8432b8551e599f24073ae : SUCCESS in 1h 03m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccd248dab6e5447bbb8a30ead448b491 : SUCCESS in 23m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a8cde92c38014b0ba0db3ac4feefb5bf : TIMED_OUT in 41m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b536902fd40f449f9b078e0f65d6325e : SUCCESS in 5m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a3618acd06c64ce193883199c3c94335 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e9141111a4e943279b0e1b94cd48f3ea : SUCCESS in 6m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/842326786cbc4fddaa6632c28cd3d5a3 : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0c84eae6ac4b4287a7f9f28bd1c93e94 : SUCCESS in 39m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/373884db124c49909fdc69ff2ac093f1 : FAILURE in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f40f818177e447679bb576f05bed0945 : SUCCESS in 1h 41m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/94cb5430ddae4b29a0e9770a2cc74760 : FAILURE in 2h 56m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/52c9ddc998654e399f7b7e3c900ef18f : SUCCESS in 1h 54m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c66457f6289941c0b4e34986d9c271b3 : SUCCESS in 2h 14m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e999db81ae2b433a90baa7f1f372eb76 : SUCCESS in 36m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b60ff9bc7e3c4e34b1b42ed30ec603b2 : SUCCESS in 1h 14m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/45e3c67bf4eb4133bec67263c693d7f7 : SUCCESS in 1h 40m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e69a612a3a194e50b9e0cc150aa4d0c9 : SUCCESS in 39m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0e3949ed3a3342ce921977ca7d544132 : SUCCESS in 2h 17m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bc7b0e0436014ccda851fe54906143cc : SUCCESS in 1h 49m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63f62df6d5034a82beb7e96d24ed511a : SUCCESS in 1h 04m 17s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":51},{"id":"61f72e3777474292a3a70add5e7bfce3f48d61e7","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-04-30 18:54:43.000000000","message":"Patch Set 51:\n\nBuild failed.\n\n- hpe-fc-driver build/66e734b4c9d24eed9c0925a6f7baf571 : FAILURE in 30m 11s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/949483/etc/tempest_log.log","accounts_in_message":[],"_revision_number":51},{"id":"180a54e7191dbe3013e74c1c9367c7aa4527a221","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-30 20:55:25.000000000","message":"Patch Set 51:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/949483 : FAILURE","accounts_in_message":[],"_revision_number":51},{"id":"7d5fefc3e86e65721f9f88ab118841be39108392","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-01 00:22:46.000000000","message":"Patch Set 51:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/949483 : SUCCESS","accounts_in_message":[],"_revision_number":51},{"id":"d2ed188f242166983e6ad1a3cd86d65f5e7abcf6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-01 00:54:55.000000000","message":"Patch Set 51:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3853/.","accounts_in_message":[],"_revision_number":51},{"id":"5bffeefa58f3f1090a6814b2323e731bf0e88144","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-01 01:03:42.000000000","message":"Patch Set 51:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4029/.","accounts_in_message":[],"_revision_number":51},{"id":"559ffb498872863c906696cf6ce311585e5a4e12","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-15 12:32:00.000000000","message":"Patch Set 51:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/f49c1918afa944d7a2dde5ac5f481772 : FAILURE in 29m 32s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/949483/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/949483/etc/tempest_log.log","accounts_in_message":[],"_revision_number":51},{"id":"72381f35d95b36f9acea6e83b1a9ef9e3137c158","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"real_author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-06-08 05:49:12.000000000","message":"Patch Set 52: Patch Set 51 was rebased on behalf of \u003cGERRIT_ACCOUNT_36538\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"}],"_revision_number":52},{"id":"e8da058c8d51998c82faa4262fa0e6dfdfbe3798","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-08 05:49:21.000000000","message":"Patch Set 52:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/854/","accounts_in_message":[],"_revision_number":52},{"id":"673c7cf6585fc8b01c9c218af0dd1367ed60030b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-08 05:49:25.000000000","message":"Patch Set 52:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/52/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/52/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":52},{"id":"0d8a5591b66513a94525d385fae24f0ff748eca7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-08 05:50:31.000000000","message":"Patch Set 52:\n\nBuild failed.\n\n- cinder-storpool-tempest build/723a2e09b0324583bfbf6f46a861b355 : NODE_FAILURE Node(set) request 200-0000000805 failed in 0s\n- cinder-storpool-tempest-iscsi-multipath build/288f6408b20844a69d27a88f5d391e0f : NODE_FAILURE Node(set) request 200-0000000806 failed in 0s","accounts_in_message":[],"_revision_number":52},{"id":"fc2c404f44ae01d5d83e70ff7bc538c331cc7653","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-06-08 05:51:04.000000000","message":"Uploaded patch set 53: New patch set was added with same tree, parent tree, and commit message as Patch Set 52.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":53},{"id":"be1d2a0e210b376a06abc2df47191c786449e2dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-08 05:51:31.000000000","message":"Patch Set 53:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/855/","accounts_in_message":[],"_revision_number":53},{"id":"3b17a760df549592bdd7be3f5aa6a313c3dd0a87","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-08 05:51:35.000000000","message":"Patch Set 53:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/53/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949483/53/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":53},{"id":"3823d9ca7980a64cd3493251d559cdd19678978f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-08 05:52:52.000000000","message":"Patch Set 53:\n\nBuild failed.\n\n- cinder-storpool-tempest build/45ef581d142e48b2a390a486cf23ca00 : NODE_FAILURE Node(set) request 200-0000000807 failed in 0s\n- cinder-storpool-tempest-iscsi-multipath build/2b8129115cc34646906b765a0fe90a24 : NODE_FAILURE Node(set) request 200-0000000808 failed in 0s","accounts_in_message":[],"_revision_number":53},{"id":"dd9a19b221ba2be84c50d81c57d1b9877a4ab527","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-08 06:15:28.000000000","message":"Patch Set 53:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2e8657bd75904f1e880c323cc75b453d\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/507bea2ac70e4f2aaf05f2f36ff5eeec : NODE_FAILURE Node request 200-0000063729 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/34899feeff704bd7904723bbc3896b04 : NODE_FAILURE Node request 200-0000063730 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/20877b58331d4e0a8554c4bba15bb4d2 : RETRY_LIMIT in 33s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ddeb9cb026c94e1c856de8a48258af84 : RETRY_LIMIT in 32s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c4a48da37d8f471483771a18f5aaccc3 : FAILURE in 22m 59s","accounts_in_message":[],"_revision_number":53},{"id":"63ac514f1c7d180f6c08745a3f92ae8496bb8492","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-08 06:17:32.000000000","message":"Patch Set 53:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/301c4675ab20446ab3f736aca136601c : FAILURE in 24m 11s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/949483","accounts_in_message":[],"_revision_number":53},{"id":"5ba96e8354b4f697ed3f1dc5ec86469d82811173","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-08 06:37:23.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- hpe-fc-driver build/834e84e283e74f3aa9e91c16f5d48570 : SUCCESS in 43m 09s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/949483","accounts_in_message":[],"_revision_number":53},{"id":"0a2cb4c554f087bb71d023918c4e3544cce855cd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-08 06:42:40.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2969fe1eeaab48e1aa512edfe1b92f28 : SUCCESS in 45m 14s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/949483\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/949483","accounts_in_message":[],"_revision_number":53},{"id":"661f2bbb14ec2ca7ff63d58e80eef80183c77471","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-08 06:46:02.000000000","message":"Patch Set 52:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":52},{"id":"2159013a28db23b9594eb58572a1775ba952fcbb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-08 06:47:40.000000000","message":"Patch Set 53:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I599167d0b63ff1d4cd0330b390a2dc33a4f06efb .","accounts_in_message":[],"_revision_number":53},{"id":"d95a4217803c838d15cdb48265fb8abb389b1cad","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-08 06:51:54.000000000","message":"Patch Set 52:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4376/.","accounts_in_message":[],"_revision_number":52},{"id":"630215c37af7baeec9c483db3226f5e4e7e0ca54","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-08 07:58:15.000000000","message":"Patch Set 53:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4377/.","accounts_in_message":[],"_revision_number":53},{"id":"14a7a29421f42dac95f787d2079d59f20a49e751","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 08:03:08.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9d30a0d496d483baee22c1dd1665cc6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01e1ed0aa91b462a95072dccbb381739 : SUCCESS in 21m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a1c3dda4d8eb4efe9c67f0aa3376b620 : SUCCESS in 17m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/678e2d2d92954774a14b0bdc67b8a0b9 : TIMED_OUT in 1h 02m 23s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"3776ee8ca1a26806601ba2be26d4e7fcc65b2217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 09:48:08.000000000","message":"Patch Set 53: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9a9662a3df34737b243b9d3097a3531\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f10e7bfc2e1435d9c7c05d178c0db71 : FAILURE in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ece9ee6cef444247a07a4cc9f011f29b : SUCCESS in 11m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c9a89a9258ac4b5ab7b5792d03b4daf1 : SUCCESS in 10m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6184a7c8340440558c2bb748c9c3ec40 : SUCCESS in 14m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde2275ae786496cbd0dca0bb8abfd3e : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/aeac6fa16c4146c3968466b9eb654941 : SUCCESS in 1h 01m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0616bb50bba44b8da0628af8e9495002 : SUCCESS in 30m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7291e69bdea84ba18827e75f16d5968b : SUCCESS in 1h 20m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51b01f8b8a7b492f9da2f83b5e29f237 : SUCCESS in 50m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b73681d3b8994d14a6e980c44ce2ae1e : SUCCESS in 9m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/59157818cd9541168471f43c9c376064 : TIMED_OUT in 42m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/909d2b97b4c64b65ad94c2a19798ba0d : SUCCESS in 2m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6878b475bfbb46f4a4fccb5cd8d3d9be : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/627913f507804edca2cc1a80df8b5e48 : SUCCESS in 4m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/33f1be1bf3dc48efbeafb411bf2fd516 : SUCCESS in 5m 51s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5c4d6d6319794d59b48313616b2e622d : SUCCESS in 6m 01s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/242454d1c2714c86ab8fa7e0edab3ce0 : SUCCESS in 28m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3916f60f9fe242e9b8894bba5131b3c5 : SUCCESS in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/070a70a118d6473f95136156fbb7b68b : FAILURE in 16m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f06e74b94ba84997b5aef470dae2b3f5 : TIMED_OUT Host unreachable in 3h 49m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c956b4e3112c441aafefd9f17e7ec7e2 : SUCCESS in 1h 58m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bf2b231dd0c74bccbb16d25a24b52636 : SUCCESS in 2h 11m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/99309a489db04a1788591183dcdec958 : SUCCESS in 24m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7c975940077749bd99d549d8fffbe264 : FAILURE in 20m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1e16e19763df4a26a47e8e451915afb5 : SUCCESS in 1h 39m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7cb373e9233c4152bf72bf28cd7083ff : SUCCESS in 41m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8d58a427d09449ea8a3bf6dffd3bce7 : SUCCESS in 1h 48m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1200c6b1c37d4d2f9a0ab445acae10c9 : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":53},{"id":"7abd16501798efcc2810645290ca3abef4abb012","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-08 10:13:30.000000000","message":"Patch Set 52:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/4a6deeb873524282913d45be8b6c1e28 : SUCCESS in 52m 57s\n- cinder-storpool-tempest-iscsi-multipath build/fabefdecedbb46e1b3476b1707c4d99d : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":52},{"id":"bc816517ea8c364618d001f4dbfa5e2b8398d8c3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-08 11:16:55.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/b6bfecaeb4184afcb42874875178bb1b : SUCCESS in 55m 15s\n- cinder-storpool-tempest-iscsi-multipath build/5d92cf5f3f0346c3812b4cb6764eab66 : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":53},{"id":"2da335020c127fe00d25bb33f06fd570ed6bdf25","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:54:46.000000000","message":"Patch Set 47:\n\nBuild Failed","accounts_in_message":[],"_revision_number":47},{"id":"00109a115eb6dabc75883716f62517d84fc30b6b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:56:49.000000000","message":"Patch Set 49:\n\nBuild Failed","accounts_in_message":[],"_revision_number":49},{"id":"b66cf88066f3441a1afe5aa22eb8e951402228a9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:58:47.000000000","message":"Patch Set 50:\n\nBuild Failed","accounts_in_message":[],"_revision_number":50},{"id":"09d31d75d7673c92b328607da8bb3b49080e826a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:02:57.000000000","message":"Patch Set 48:\n\nBuild Failed","accounts_in_message":[],"_revision_number":48},{"id":"da196f4f0e08447447a1ee40e56a5f6313c82f17","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:03:11.000000000","message":"Patch Set 51:\n\nBuild Failed","accounts_in_message":[],"_revision_number":51},{"id":"afbdbdfb29b5e1043cf03bfd46e9626eedcabe2c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-08 03:19:08.000000000","message":"Patch Set 52:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4200/.","accounts_in_message":[],"_revision_number":52},{"id":"819d007e75520958b3164ba8cd1d333c9e1aa030","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-08 04:22:40.000000000","message":"Patch Set 53:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4201/.","accounts_in_message":[],"_revision_number":53},{"id":"8e7bd7eb20c030107ebe297ef5344509810c0f51","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":39312,"name":"Nimesh Desai","display_name":"Nimesh","email":"talk2nimeshdesai@gmail.com"},"date":"2026-07-13 07:43:27.000000000","message":"Hashtag added: ibm-svf-cinder","accounts_in_message":[],"_revision_number":53}],"current_revision_number":53,"current_revision":"26ac9811793a421c565d0dc4527e69fd3d675d5f","revisions":{"4059a067f698a7df3456b65a41e189f7e3b33777":{"kind":"REWORK","_number":1,"created":"2025-05-12 10:24:34.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/83/949483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/1"}}},"commit":{"parents":[{"commit":"b856589e7e79fc45f223abfa94959971924ed895","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b856589e7e79fc45f223abfa94959971924ed895"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-12 10:20:29.000000000","tz":330},"committer":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-12 10:20:29.000000000","tz":330},"subject":"[IBM SVf driver] Adding support for PBHA partition backend","message":"[IBM SVf driver] Adding support for PBHA partition backend\n\n    [Storage Virtualize family] As part of PBHA implementation,\n    added support to extend a PBHA volume in cinder driver.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4059a067f698a7df3456b65a41e189f7e3b33777"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4059a067f698a7df3456b65a41e189f7e3b33777"}]},"branch":"refs/heads/master"},"92cdff4040c5a137cdad250530a2fcc87a51644f":{"kind":"REWORK","_number":2,"created":"2025-05-20 07:25:28.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/83/949483/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/2"}}},"commit":{"parents":[{"commit":"b856589e7e79fc45f223abfa94959971924ed895","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b856589e7e79fc45f223abfa94959971924ed895"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-12 10:20:29.000000000","tz":330},"committer":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 07:16:48.000000000","tz":330},"subject":"[IBM SVf driver] Adding support for PBHA partition backend","message":"[IBM SVf driver] Adding support for PBHA partition backend\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume in cinder driver.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92cdff4040c5a137cdad250530a2fcc87a51644f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92cdff4040c5a137cdad250530a2fcc87a51644f"}]},"branch":"refs/heads/master"},"68cb1da53d454efd6fd3f792ced87e7a9757f16c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-05-20 07:34:24.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/83/949483/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/3"}}},"commit":{"parents":[{"commit":"b856589e7e79fc45f223abfa94959971924ed895","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b856589e7e79fc45f223abfa94959971924ed895"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-12 10:20:29.000000000","tz":330},"committer":{"name":"Akshay Ranade","email":"akshay.ranade@ibm.com","date":"2025-05-20 07:34:24.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume in cinder driver.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68cb1da53d454efd6fd3f792ced87e7a9757f16c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68cb1da53d454efd6fd3f792ced87e7a9757f16c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9adeb1a00347895e9a1f5f59b0bb94ef4a42e7af":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-05-20 07:35:06.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/83/949483/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/4"}}},"commit":{"parents":[{"commit":"b856589e7e79fc45f223abfa94959971924ed895","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b856589e7e79fc45f223abfa94959971924ed895"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-12 10:20:29.000000000","tz":330},"committer":{"name":"Akshay Ranade","email":"akshay.ranade@ibm.com","date":"2025-05-20 07:35:06.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9adeb1a00347895e9a1f5f59b0bb94ef4a42e7af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9adeb1a00347895e9a1f5f59b0bb94ef4a42e7af"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5af02a943a5c74f1af9e597475e59b75d898f7b8":{"kind":"REWORK","_number":5,"created":"2025-05-20 09:22:22.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/83/949483/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/5"}}},"commit":{"parents":[{"commit":"d23a17da679e1180a546d78138e257f397fe4b4b","subject":"Merge \"nfs: minimal mypy coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d23a17da679e1180a546d78138e257f397fe4b4b"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5af02a943a5c74f1af9e597475e59b75d898f7b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5af02a943a5c74f1af9e597475e59b75d898f7b8"}]},"branch":"refs/heads/master"},"9287ff4279cf6ada74037a0473b44125ae65b4d1":{"kind":"REWORK","_number":6,"created":"2025-05-20 15:39:22.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/83/949483/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/6"}}},"commit":{"parents":[{"commit":"d23a17da679e1180a546d78138e257f397fe4b4b","subject":"Merge \"nfs: minimal mypy coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d23a17da679e1180a546d78138e257f397fe4b4b"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 15:38:13.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9287ff4279cf6ada74037a0473b44125ae65b4d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9287ff4279cf6ada74037a0473b44125ae65b4d1"}]},"branch":"refs/heads/master"},"71c168453715f355c932771f8a3a4e49bed5eb03":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-05-26 12:14:54.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/7"}}},"commit":{"parents":[{"commit":"b856589e7e79fc45f223abfa94959971924ed895","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b856589e7e79fc45f223abfa94959971924ed895"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-05-26 12:14:03.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71c168453715f355c932771f8a3a4e49bed5eb03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71c168453715f355c932771f8a3a4e49bed5eb03"}]},"branch":"refs/heads/master"},"b8da0a6bc7054dc0b901e7f928e5d66ff33a4582":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-05-30 06:53:29.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"real_uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/83/949483/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/8"}}},"commit":{"parents":[{"commit":"6f5c45f0bc11ff87a1ad1831005d9b4812b59572","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f5c45f0bc11ff87a1ad1831005d9b4812b59572"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Vrinda Dhakad","email":"vrinda.dhakad@ibm.com","date":"2025-05-30 06:53:29.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8da0a6bc7054dc0b901e7f928e5d66ff33a4582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8da0a6bc7054dc0b901e7f928e5d66ff33a4582"}]},"branch":"refs/heads/master","description":"Rebase"},"c9302564d7cdb79216d1312419612fdd98f57742":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-06-05 12:08:59.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"real_uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/9"}}},"commit":{"parents":[{"commit":"75356116e3e4dd7061bb27fd20717070947a2748","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75356116e3e4dd7061bb27fd20717070947a2748"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Vrinda Dhakad","email":"vrinda.dhakad@ibm.com","date":"2025-06-05 12:08:59.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9302564d7cdb79216d1312419612fdd98f57742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9302564d7cdb79216d1312419612fdd98f57742"}]},"branch":"refs/heads/master","description":"Rebase"},"794971d2d26ffb550ea8e0302f6c6154939082f0":{"kind":"NO_CHANGE","_number":10,"created":"2025-06-05 12:17:59.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/10"}}},"commit":{"parents":[{"commit":"75356116e3e4dd7061bb27fd20717070947a2748","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75356116e3e4dd7061bb27fd20717070947a2748"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-05 12:10:46.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/794971d2d26ffb550ea8e0302f6c6154939082f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/794971d2d26ffb550ea8e0302f6c6154939082f0"}]},"branch":"refs/heads/master"},"d4c7043be2308d1d82040ff84c7bda06d6117cbc":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-06-05 19:15:24.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/11"}}},"commit":{"parents":[{"commit":"efe3ef6fc00db181ca876a7de0f745a57b2de13e","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efe3ef6fc00db181ca876a7de0f745a57b2de13e"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-05 19:15:24.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4c7043be2308d1d82040ff84c7bda06d6117cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4c7043be2308d1d82040ff84c7bda06d6117cbc"}]},"branch":"refs/heads/master","description":"Rebase"},"b9c706fa0acfe7fa958a4f962743193276bc88ff":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-06-06 06:33:50.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/12"}}},"commit":{"parents":[{"commit":"34d06403e3c483b56f442cd8e6922d05da83c5e3","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34d06403e3c483b56f442cd8e6922d05da83c5e3"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2025-06-06 06:33:50.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9c706fa0acfe7fa958a4f962743193276bc88ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9c706fa0acfe7fa958a4f962743193276bc88ff"}]},"branch":"refs/heads/master","description":"Rebase"},"71177d82aec2798577194c21181a13eddce7c0d4":{"kind":"NO_CHANGE","_number":13,"created":"2025-06-09 10:53:50.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/13"}}},"commit":{"parents":[{"commit":"a8c03be5ab93caf95baa3255930554e0530792ee","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8c03be5ab93caf95baa3255930554e0530792ee"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-06-09 10:40:36.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71177d82aec2798577194c21181a13eddce7c0d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71177d82aec2798577194c21181a13eddce7c0d4"}]},"branch":"refs/heads/master"},"411a6d6bd9645a6d4aba13b4c5caca5e23b64ba3":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-06-12 08:53:15.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/14"}}},"commit":{"parents":[{"commit":"c80d541676fe0532092b4febf6339618e3d9f2cd","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c80d541676fe0532092b4febf6339618e3d9f2cd"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-06-11 15:49:45.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/411a6d6bd9645a6d4aba13b4c5caca5e23b64ba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/411a6d6bd9645a6d4aba13b4c5caca5e23b64ba3"}]},"branch":"refs/heads/master"},"f928316825fdbae69cffd923281bdbcc68dd0e55":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":15,"created":"2025-07-08 08:33:46.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/15"}}},"commit":{"parents":[{"commit":"3b3dfceb5df55018a439cba66be45bb9af04c28a","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3b3dfceb5df55018a439cba66be45bb9af04c28a"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-07-08 08:33:01.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f928316825fdbae69cffd923281bdbcc68dd0e55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f928316825fdbae69cffd923281bdbcc68dd0e55"}]},"branch":"refs/heads/master"},"ad737f5009eb868d4ae290ecf5f8d0775f03ee35":{"kind":"NO_CHANGE","_number":16,"created":"2025-07-17 10:07:33.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"real_uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/16"}}},"commit":{"parents":[{"commit":"118d021badfe818f4fc0a2ea17a4666d7d41291c","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/118d021badfe818f4fc0a2ea17a4666d7d41291c"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Vrinda Dhakad","email":"vrinda.dhakad@ibm.com","date":"2025-07-17 10:07:33.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad737f5009eb868d4ae290ecf5f8d0775f03ee35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad737f5009eb868d4ae290ecf5f8d0775f03ee35"}]},"branch":"refs/heads/master","description":"Rebase"},"8f9e993d2ec6e52187d511e21eedcd95e6842794":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2025-08-21 03:32:09.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/17"}}},"commit":{"parents":[{"commit":"dfdac06f061a08ae9d80d6a6da7062989b73a0f1","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dfdac06f061a08ae9d80d6a6da7062989b73a0f1"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-07-30 07:00:24.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f9e993d2ec6e52187d511e21eedcd95e6842794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f9e993d2ec6e52187d511e21eedcd95e6842794"}]},"branch":"refs/heads/master"},"d01ae920b78ab9a5c1315194c22fdc4b645908ce":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2025-09-10 08:34:13.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/18"}}},"commit":{"parents":[{"commit":"fdf87a280a0d1a983e7c17e66ca6c7a3465e1095","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fdf87a280a0d1a983e7c17e66ca6c7a3465e1095"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Vrinda Dhakad","email":"vrinda.dhakad@ibm.com","date":"2025-09-10 08:34:13.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d01ae920b78ab9a5c1315194c22fdc4b645908ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d01ae920b78ab9a5c1315194c22fdc4b645908ce"}]},"branch":"refs/heads/master","description":"Rebase"},"4fb57f801af17f19bd63b6eddb92e0c08a175f49":{"kind":"NO_CHANGE","_number":19,"created":"2025-09-10 08:37:51.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/19"}}},"commit":{"parents":[{"commit":"fdf87a280a0d1a983e7c17e66ca6c7a3465e1095","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fdf87a280a0d1a983e7c17e66ca6c7a3465e1095"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-09-10 08:33:55.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4fb57f801af17f19bd63b6eddb92e0c08a175f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4fb57f801af17f19bd63b6eddb92e0c08a175f49"}]},"branch":"refs/heads/master"},"f5b5ee65e5a250cb889931a236368b7004186e50":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2025-09-29 06:04:21.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/20"}}},"commit":{"parents":[{"commit":"c1e745cd43220293eb994a597f3f7158df620253","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1e745cd43220293eb994a597f3f7158df620253"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Vrinda Dhakad","email":"vrinda.dhakad@ibm.com","date":"2025-09-29 06:04:21.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5b5ee65e5a250cb889931a236368b7004186e50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5b5ee65e5a250cb889931a236368b7004186e50"}]},"branch":"refs/heads/master","description":"Rebase"},"d1803618817afce39f810587d161b3ed338f8b09":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":21,"created":"2025-10-09 19:06:48.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/83/949483/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/21"}}},"commit":{"parents":[{"commit":"97bb9db9181e91e542deb3117489674ce6bdbaca","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97bb9db9181e91e542deb3117489674ce6bdbaca"}]}],"author":{"name":"Akshay Ranade","email":"Akshay.Ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-10-09 19:03:19.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1803618817afce39f810587d161b3ed338f8b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1803618817afce39f810587d161b3ed338f8b09"}]},"branch":"refs/heads/master"},"c8fbcb2cca8740e492d18805eb207bb8e4037488":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2025-10-16 07:08:57.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/22"}}},"commit":{"parents":[{"commit":"7e82780b232ba033f95d13de98b0fb3327c12028","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e82780b232ba033f95d13de98b0fb3327c12028"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-10-16 06:58:05.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8fbcb2cca8740e492d18805eb207bb8e4037488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8fbcb2cca8740e492d18805eb207bb8e4037488"}]},"branch":"refs/heads/master"},"55b444162257766a32c98beaa3cfbc53f1f23925":{"kind":"NO_CHANGE","_number":23,"created":"2025-10-17 07:18:43.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/83/949483/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/23"}}},"commit":{"parents":[{"commit":"4b98f26dc7b63b4ce036e8b39173caed0a8f4fa3","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b98f26dc7b63b4ce036e8b39173caed0a8f4fa3"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-10-17 06:23:19.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55b444162257766a32c98beaa3cfbc53f1f23925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55b444162257766a32c98beaa3cfbc53f1f23925"}]},"branch":"refs/heads/master"},"93ed542a484460c3c48bfcd9f177edc8f99ace73":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2025-10-31 06:57:42.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/24"}}},"commit":{"parents":[{"commit":"c151db4d06ee896a411c93acd9e8771f3e904668","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c151db4d06ee896a411c93acd9e8771f3e904668"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-10-31 06:27:39.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/93ed542a484460c3c48bfcd9f177edc8f99ace73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/93ed542a484460c3c48bfcd9f177edc8f99ace73"}]},"branch":"refs/heads/master"},"494cae803c1c93de35eb7b5614fd5ca55c0f08e2":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2025-10-31 07:07:32.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/25"}}},"commit":{"parents":[{"commit":"3198bcd499161b72e336c105647fe0a54bf5591d","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3198bcd499161b72e336c105647fe0a54bf5591d"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-10-31 07:06:30.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/494cae803c1c93de35eb7b5614fd5ca55c0f08e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/494cae803c1c93de35eb7b5614fd5ca55c0f08e2"}]},"branch":"refs/heads/master"},"99054743406acaba33f04fed3d32335d48d2e718":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2025-10-31 15:32:34.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/83/949483/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/26"}}},"commit":{"parents":[{"commit":"bfe4ad5cfd4025d108b224b04b16566baf4c1a4c","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bfe4ad5cfd4025d108b224b04b16566baf4c1a4c"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"dakshinapurohit","email":"Dakshina.Purohit@ibm.com","date":"2025-10-31 20:56:18.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99054743406acaba33f04fed3d32335d48d2e718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99054743406acaba33f04fed3d32335d48d2e718"}]},"branch":"refs/heads/master"},"b04b9c4abce59ec168afa9c7ed1ee65c6c9e0a94":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2025-11-11 10:20:43.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/27"}}},"commit":{"parents":[{"commit":"c9ed58d17cb3093407009ca9979dfbc06ee47e9d","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9ed58d17cb3093407009ca9979dfbc06ee47e9d"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-11-11 10:16:22.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b04b9c4abce59ec168afa9c7ed1ee65c6c9e0a94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b04b9c4abce59ec168afa9c7ed1ee65c6c9e0a94"}]},"branch":"refs/heads/master"},"c85e0e3157a11440323f577529eee540d4ab752e":{"kind":"REWORK","_number":28,"created":"2025-11-12 09:35:16.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/83/949483/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/28"}}},"commit":{"parents":[{"commit":"40873cc96825b9013c0b1580141e0be29e260eaa","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40873cc96825b9013c0b1580141e0be29e260eaa"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-11-12 09:34:30.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c85e0e3157a11440323f577529eee540d4ab752e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c85e0e3157a11440323f577529eee540d4ab752e"}]},"branch":"refs/heads/master"},"6c91b4d431e28fa78d827de1eb2503af206246f0":{"kind":"REWORK","_number":29,"created":"2025-11-12 13:11:35.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/83/949483/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/29"}}},"commit":{"parents":[{"commit":"40873cc96825b9013c0b1580141e0be29e260eaa","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40873cc96825b9013c0b1580141e0be29e260eaa"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-11-12 13:10:24.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c91b4d431e28fa78d827de1eb2503af206246f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c91b4d431e28fa78d827de1eb2503af206246f0"}]},"branch":"refs/heads/master"},"227b5cd0b63f0d613d6523dfadea935cf409c27d":{"kind":"REWORK","_number":30,"created":"2025-11-13 10:22:02.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/83/949483/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/30"}}},"commit":{"parents":[{"commit":"48bf981b916dd2788dc84276927dcc126d7e927a","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48bf981b916dd2788dc84276927dcc126d7e927a"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-11-13 10:21:35.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/227b5cd0b63f0d613d6523dfadea935cf409c27d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/227b5cd0b63f0d613d6523dfadea935cf409c27d"}]},"branch":"refs/heads/master"},"9a372c31f94e4374368586ea1273bd225b2d914f":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2025-11-13 11:52:09.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/83/949483/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/31"}}},"commit":{"parents":[{"commit":"50fcd5ac12a82b5572ec97e2720424a74393053d","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50fcd5ac12a82b5572ec97e2720424a74393053d"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-11-13 11:32:07.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a372c31f94e4374368586ea1273bd225b2d914f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a372c31f94e4374368586ea1273bd225b2d914f"}]},"branch":"refs/heads/master"},"1c0e911f1dd490877b044ade5bdfdb7ffadcbd45":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2025-11-13 11:57:28.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/83/949483/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/32"}}},"commit":{"parents":[{"commit":"53a75c8770cf6e9879382c53b1ade4f3535dfb02","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53a75c8770cf6e9879382c53b1ade4f3535dfb02"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"dakshinapurohit","email":"Dakshina.Purohit@ibm.com","date":"2025-11-13 16:50:55.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c0e911f1dd490877b044ade5bdfdb7ffadcbd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c0e911f1dd490877b044ade5bdfdb7ffadcbd45"}]},"branch":"refs/heads/master"},"c2b25224116175df99d8d5a7d8c109ef28f5499d":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2025-11-20 05:58:07.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/33"}}},"commit":{"parents":[{"commit":"ce594e11f0cc9c6bb8f10dc4f8e903e3e5232253","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce594e11f0cc9c6bb8f10dc4f8e903e3e5232253"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-11-20 05:58:07.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2b25224116175df99d8d5a7d8c109ef28f5499d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2b25224116175df99d8d5a7d8c109ef28f5499d"}]},"branch":"refs/heads/master","description":"Rebase"},"9793a55aae3449048a711dae53d63165f6910770":{"kind":"NO_CHANGE","_number":34,"created":"2025-11-20 08:48:14.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/34"}}},"commit":{"parents":[{"commit":"ce594e11f0cc9c6bb8f10dc4f8e903e3e5232253","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce594e11f0cc9c6bb8f10dc4f8e903e3e5232253"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-11-20 06:08:40.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9793a55aae3449048a711dae53d63165f6910770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9793a55aae3449048a711dae53d63165f6910770"}]},"branch":"refs/heads/master"},"3f9e1da1ae0a4640e154df75d0e1d5f3b32c6c98":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2026-01-05 08:30:10.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/83/949483/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/35"}}},"commit":{"parents":[{"commit":"b4d68e13bb0934e8eeafc9771582f3fc617ed27e","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4d68e13bb0934e8eeafc9771582f3fc617ed27e"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-12-24 08:52:53.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f9e1da1ae0a4640e154df75d0e1d5f3b32c6c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f9e1da1ae0a4640e154df75d0e1d5f3b32c6c98"}]},"branch":"refs/heads/master"},"407ca739b15cbc33f75fd8e4c2308ef8ab811613":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2026-01-06 10:23:31.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/36"}}},"commit":{"parents":[{"commit":"f47e00ba9b5cb312cad716417954adab13e39db4","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f47e00ba9b5cb312cad716417954adab13e39db4"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-01-06 09:13:32.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/407ca739b15cbc33f75fd8e4c2308ef8ab811613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/407ca739b15cbc33f75fd8e4c2308ef8ab811613"}]},"branch":"refs/heads/master"},"1266c8c185b21d5b987bb8c85b47f7aec14f8760":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2026-01-22 05:37:55.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"real_uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/37"}}},"commit":{"parents":[{"commit":"689a7046dca42520528df8cada5675a37bbfbe94","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/689a7046dca42520528df8cada5675a37bbfbe94"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2026-01-22 05:37:55.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1266c8c185b21d5b987bb8c85b47f7aec14f8760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1266c8c185b21d5b987bb8c85b47f7aec14f8760"}]},"branch":"refs/heads/master","description":"Rebase"},"2cc9c76c90d8338a07315dbbdfc08dbbb19d58b3":{"kind":"NO_CHANGE","_number":38,"created":"2026-01-22 06:04:03.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/38"}}},"commit":{"parents":[{"commit":"689a7046dca42520528df8cada5675a37bbfbe94","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/689a7046dca42520528df8cada5675a37bbfbe94"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-20 05:56:07.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2cc9c76c90d8338a07315dbbdfc08dbbb19d58b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2cc9c76c90d8338a07315dbbdfc08dbbb19d58b3"}]},"branch":"refs/heads/master"},"50dca409d363a654915de40f31d8b15760b5b48e":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2026-01-28 06:59:48.000000000","uploader":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"ref":"refs/changes/83/949483/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/39"}}},"commit":{"parents":[{"commit":"7bd67c1886542f40b28a62b9e789efa8d2400914","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7bd67c1886542f40b28a62b9e789efa8d2400914"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Sandip27","email":"sandip.rajbanshi@ibm.com","date":"2026-01-27 10:58:09.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50dca409d363a654915de40f31d8b15760b5b48e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50dca409d363a654915de40f31d8b15760b5b48e"}]},"branch":"refs/heads/master"},"6bfd6f10da8ca989ce274fbb2618084f6a87805b":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2026-01-29 11:51:22.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/83/949483/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/40"}}},"commit":{"parents":[{"commit":"362292c825b28b6b237e4bdae8e4fc54b085a2b6","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/362292c825b28b6b237e4bdae8e4fc54b085a2b6"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-01-29 11:46:06.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6bfd6f10da8ca989ce274fbb2618084f6a87805b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6bfd6f10da8ca989ce274fbb2618084f6a87805b"}]},"branch":"refs/heads/master"},"806172db0bed752e7520cf67efefce1e1e2bfeac":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2026-02-05 06:38:56.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/83/949483/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/41"}}},"commit":{"parents":[{"commit":"2e4842de7121a726051f7091faf7219590aa821e","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e4842de7121a726051f7091faf7219590aa821e"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-02-05 06:35:41.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/806172db0bed752e7520cf67efefce1e1e2bfeac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/806172db0bed752e7520cf67efefce1e1e2bfeac"}]},"branch":"refs/heads/master"},"b364084bb727e46138997c5fb1fcc33c92db0182":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2026-02-05 06:52:25.000000000","uploader":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"ref":"refs/changes/83/949483/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/42"}}},"commit":{"parents":[{"commit":"66b25db86635592dec3637152eb53e4540445fc8","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/66b25db86635592dec3637152eb53e4540445fc8"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"sandip-rajbanshi-27","email":"sandip.rajbanshi@ibm.com","date":"2026-02-05 06:50:40.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b364084bb727e46138997c5fb1fcc33c92db0182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b364084bb727e46138997c5fb1fcc33c92db0182"}]},"branch":"refs/heads/master"},"5bb6b862e344eff5b7d9426e1ca25de3af5d9415":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2026-02-05 09:44:17.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/43"}}},"commit":{"parents":[{"commit":"d1b4e7ce8800a33e2bd89355f229afd3c25d84fb","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1b4e7ce8800a33e2bd89355f229afd3c25d84fb"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-02-05 09:34:45.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bb6b862e344eff5b7d9426e1ca25de3af5d9415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bb6b862e344eff5b7d9426e1ca25de3af5d9415"}]},"branch":"refs/heads/master"},"cfd38d91e7ec3065214025714f8724a753d891f0":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2026-02-13 09:22:24.000000000","uploader":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"ref":"refs/changes/83/949483/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/44"}}},"commit":{"parents":[{"commit":"2190c396975814b59051690ebbb6a31759eb51f6","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2190c396975814b59051690ebbb6a31759eb51f6"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"sandip-rajbanshi-27","email":"sandip.rajbanshi@ibm.com","date":"2026-02-13 07:39:26.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cfd38d91e7ec3065214025714f8724a753d891f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cfd38d91e7ec3065214025714f8724a753d891f0"}]},"branch":"refs/heads/master"},"efc6969341257e42941c5225fc2df28690f7841a":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2026-03-31 06:18:14.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/45"}}},"commit":{"parents":[{"commit":"a1c13111bb9d60ff685c38d3016b00a0156076ce","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1c13111bb9d60ff685c38d3016b00a0156076ce"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-03-31 05:37:53.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efc6969341257e42941c5225fc2df28690f7841a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efc6969341257e42941c5225fc2df28690f7841a"}]},"branch":"refs/heads/master"},"c4ca1cf861f2ff09bd9699af97b0d11e34a41c20":{"kind":"NO_CHANGE","_number":46,"created":"2026-03-31 09:40:48.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"real_uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/46"}}},"commit":{"parents":[{"commit":"984605391cf0a647483e9c2466b6915fc5525646","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/984605391cf0a647483e9c2466b6915fc5525646"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Vrinda Dhakad","email":"vrinda.dhakad@ibm.com","date":"2026-03-31 09:40:48.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c4ca1cf861f2ff09bd9699af97b0d11e34a41c20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c4ca1cf861f2ff09bd9699af97b0d11e34a41c20"}]},"branch":"refs/heads/master","description":"Rebase"},"d5735e28fc3d65c10345634bc24e9e821b3ee775":{"kind":"TRIVIAL_REBASE","_number":47,"created":"2026-04-09 10:19:31.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/47"}}},"commit":{"parents":[{"commit":"6f740132fee19833e8b2763bcf18943310bf62f2","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f740132fee19833e8b2763bcf18943310bf62f2"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-04-09 06:26:00.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5735e28fc3d65c10345634bc24e9e821b3ee775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5735e28fc3d65c10345634bc24e9e821b3ee775"}]},"branch":"refs/heads/master"},"1a737aaed1e7c2c41edf888122dd95276ce6be86":{"kind":"NO_CHANGE","_number":48,"created":"2026-04-13 07:27:00.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/83/949483/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/48"}}},"commit":{"parents":[{"commit":"b83efaf97e2ce5e37ca910d9d915878d91b34fe6","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b83efaf97e2ce5e37ca910d9d915878d91b34fe6"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-04-13 05:59:44.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a737aaed1e7c2c41edf888122dd95276ce6be86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a737aaed1e7c2c41edf888122dd95276ce6be86"}]},"branch":"refs/heads/master"},"4d7630ae752f88fd3548b93a7f6e83d4e00df74c":{"kind":"NO_CHANGE","_number":49,"created":"2026-04-14 11:04:51.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/83/949483/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/49"}}},"commit":{"parents":[{"commit":"4f0c64d891ffe6d8554f55f83947e245d82891dc","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f0c64d891ffe6d8554f55f83947e245d82891dc"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-04-14 11:01:04.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d7630ae752f88fd3548b93a7f6e83d4e00df74c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d7630ae752f88fd3548b93a7f6e83d4e00df74c"}]},"branch":"refs/heads/master"},"c88a5d2e6f3f8ebff62273d31dc6ff94f2a81ce3":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2026-04-28 17:53:21.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/83/949483/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/50"}}},"commit":{"parents":[{"commit":"7b612c1b3dc5dd70b6460a8539212883ff08b46a","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b612c1b3dc5dd70b6460a8539212883ff08b46a"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-04-28 17:22:41.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c88a5d2e6f3f8ebff62273d31dc6ff94f2a81ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c88a5d2e6f3f8ebff62273d31dc6ff94f2a81ce3"}]},"branch":"refs/heads/master"},"12414dfe7ce4e51abcbc54ebb379a7051335413a":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2026-04-30 15:08:44.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/83/949483/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/51"}}},"commit":{"parents":[{"commit":"3c79700387d162077eef8bcc7f6298a952089cfa","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c79700387d162077eef8bcc7f6298a952089cfa"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-04-27 18:41:34.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12414dfe7ce4e51abcbc54ebb379a7051335413a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12414dfe7ce4e51abcbc54ebb379a7051335413a"}]},"branch":"refs/heads/master"},"a5d025f7e02cd4c3243b7f184ed5a3f6fd0bafba":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2026-06-08 05:49:12.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"real_uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/52"}}},"commit":{"parents":[{"commit":"64f5cb97fa830109e40e4b2eca333f5cef0a290c","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64f5cb97fa830109e40e4b2eca333f5cef0a290c"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2026-06-08 05:49:12.000000000","tz":0},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a5d025f7e02cd4c3243b7f184ed5a3f6fd0bafba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a5d025f7e02cd4c3243b7f184ed5a3f6fd0bafba"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3c79700387d162077eef8bcc7f6298a952089cfa","ours":"12414dfe7ce4e51abcbc54ebb379a7051335413a","theirs":"64f5cb97fa830109e40e4b2eca333f5cef0a290c","merge_strategy":"recursive","contains_conflicts":false}},"26ac9811793a421c565d0dc4527e69fd3d675d5f":{"kind":"NO_CHANGE","_number":53,"created":"2026-06-08 05:51:04.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/83/949483/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/949483/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/949483/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/949483/53"}}},"commit":{"parents":[{"commit":"64f5cb97fa830109e40e4b2eca333f5cef0a290c","subject":"[IBM SVf driver] Adding support for PBHA partition backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64f5cb97fa830109e40e4b2eca333f5cef0a290c"}]}],"author":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-05-20 09:20:54.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-06-03 06:53:51.000000000","tz":330},"subject":"[IBM SVf driver] PBHA Volume Extend Support","message":"[IBM SVf driver] PBHA Volume Extend Support\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support to extend a PBHA volume.\n\nChange-Id: I599167d0b63ff1d4cd0330b390a2dc33a4f06efb\nSigned-off-by: rakshay \u003cakshay.ranade@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26ac9811793a421c565d0dc4527e69fd3d675d5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26ac9811793a421c565d0dc4527e69fd3d675d5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
