)]}'
{"id":"openstack%2Fcinder~949717","triplet_id":"openstack%2Fcinder~master~I6e2551f087b589c3e3e094f70169360d74099e57","project":"openstack/cinder","branch":"master","attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-08-27 21:00:55.000000000","reason":"Someone else replied on the change"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-08-27 11:59:33.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6e2551f087b589c3e3e094f70169360d74099e57","subject":"Avoid deep copying of volume when it\u0027s on the same filesystem","status":"NEW","created":"2025-05-14 09:56:05.000000000","updated":"2026-09-27 17:10:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"d328d7a5c6ba2f66b13eb64e862ad2c1f6e5ae8f","_number":949717,"virtual_id_number":949717,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-25 11:40:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"all":[{"value":-1,"date":"2026-09-27 17:10:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-14 10:36:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-14 10:40:56.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-05-14 10:42:50.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-14 11:01:20.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-05-14 11:27:52.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-14 12:09:08.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-05-14 12:17:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-14 12:20:39.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-14 12:26:32.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-05-14 17:25:15.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-05-15 01:58:00.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-05-16 02:23:42.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-27 12:34:54.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2026-09-25 11:40:20.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-27 17:10:51.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"}],"messages":[{"id":"e839bb2d06f2b6e3b7c6b683c75852deb808c475","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-14 09:56:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8844ba199f0e8199288c30323899482b05bc23fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-14 09:58:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"090cb1f125e3dfaa51b1e9601cb59d8a95b47d63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 10:36:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/666c47fba89942749307af58117fbd57\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/195379d491494fc1b80c5080dd8a5ea7 : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f4a2206f980e423693c3bf9def3ccca1 : SUCCESS in 20m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cf4ecb604178456bb924ac22607a613b : SUCCESS in 25m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c784d33e3df189479d9457f1277291359b4a5deb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-14 10:40:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949717 : FAILURE in 37m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f13e96ee22d9b646581985a5e02fc94c3e415c74","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-14 10:42:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/949717 : FAILURE in 37m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b4678d4ed3951b9e0ca21602906d975278f8c445","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-14 11:01:20.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/726/.","accounts_in_message":[],"_revision_number":2},{"id":"2de3bd0631ff0428b7d81ef929e3c5ae3c72b537","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-14 11:03:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/848/.","accounts_in_message":[],"_revision_number":2},{"id":"8aa8a56865631915a71515850e99c584bcad9a93","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-14 11:27:52.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"113b7d8216a384e8f770b4ef81b077b04e744465","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-14 12:09:08.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8d9e3907c8c14a6eb41297e8e1a569c0\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/88848b35ecb04f79b8f6942be41533fc : SUCCESS in 40m 41s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3aa41a29e4644b7da6ac6baefef14708 : SUCCESS in 45m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/171d54faadbb43abb8268b7667012d7f : FAILURE in 2h 03m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ff1ee45e45754244b144e27db5ab1098 : SUCCESS in 40m 41s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/464fa7271a9e42ca9a05bf7f1aa59740 : FAILURE in 46m 33s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b89fb7cddbc24da8a712a8e3a6a3ac21 : FAILURE in 57m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6b965dc6b2a64d089c6e5029aad11d39 : FAILURE in 1h 53m 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e5ecc4e24ce148b4900868824f85789b : SUCCESS in 41m 40s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#68\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#416\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#82\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#416\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#165\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job va... (truncated)","accounts_in_message":[],"_revision_number":2},{"id":"9a5b00e5aa587e0c487b42b673ae34f5fe055045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 12:17:40.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33dca106e41f41d19bcb6a83f68e38a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf2e8f0fb5c43c79a6bb9c0a0e1fe08 : FAILURE in 7m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/297744bd72624dc6a54558a9f16b09e3 : SUCCESS in 13m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7cc61747254443058b36daa21a68a8b4 : SUCCESS in 14m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9ce34c9f40048a8a2139a0e127e9297 : SUCCESS in 15m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c31fc39f66d47479e9b811f5d2940a9 : SUCCESS in 15m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/fb31e51591c64d2c8631b551c723115b : SUCCESS in 54m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/47f5fde9cd5f473692471d8280a70b0f : SUCCESS in 59m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c7b2d4b66ff0408b9355a415e62b0d93 : SUCCESS in 1h 36m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68c10ffcb467418aa21f94e136aa912e : SUCCESS in 1h 00m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/13f9b876dce742beb750a11bb854cc88 : FAILURE in 36m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a1ddf76381f14b7c8e0707d71ab68e4c : FAILURE in 7m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b7b533d35eaf47b1bc227338e7dbf722 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9be34bc59788417c9c16f248b4d406a4 : SUCCESS in 6m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/aaf65870ff3442dba7aff8f1f5cb6a74 : SUCCESS in 6m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8aaa4ee4c4b5428992b1a5c9093fbc49 : FAILURE in 31m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db4a792956474bf5b2eca9f88aec7488 : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2dc1b0a246f3492598b78b31e448b81a : FAILURE in 1h 36m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3992ac03112946e995201f467161b744 : FAILURE in 2h 12m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/181ed8c764754004bb90e1c53bc2ba1c : SUCCESS in 2h 03m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a94565f84c5547379ab06e5464781db8 : FAILURE in 11m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b4154b48f431404f9fe70146016b31cd : SUCCESS in 18m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cbb94306dbe044fba8df774391180e44 : SUCCESS in 32m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6b2d813f72944b6eb2d6ac6a29f20a6c : SUCCESS in 1h 31m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/75c79f7f73ed4ed3af775fdc74d46074 : SUCCESS in 30m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7155246353f34c858d4f0faebcfededf : SUCCESS in 1h 28m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e5edaf5c8554454d909903de095b4a38 : FAILURE in 22m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa3f0ff20f094cbb98f00dea85e3bafb : SUCCESS in 1h 49m 18s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3794fb44982b4d2d9455a6f003a9d307 : SUCCESS in 1h 30m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb1f3a6ba0b444cf9c4a7398eaf1f2ff : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4c990f786acd0331498f36e293a229e903aae413","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-14 12:20:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/65c11fc021df47b8a9f77d2efab69c3e : SUCCESS in 1h 13m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e8ea155273b84749a0dc805814c99b12 : SUCCESS in 1h 16m 47s","accounts_in_message":[],"_revision_number":2},{"id":"3e6dfdabe9dbf83f35535d70bd8a4ff9733b9575","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-05-14 12:26:32.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e117ddf15f4dc1abdf7d37a54739e7bc3a3bd222","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-14 13:05:52.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7209dde3626ed1e7d1073a7da5b1e54b3a81267c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-05-14 17:25:15.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-949717-1/FC/ : FAILURE in 7h 28m 44s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-949717-1/iSCSI/ : SUCCESS in 1h 47m 38s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"e66fa09a79a69ef2f9649fa7d3e1cc0b09ae3698","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-15 01:58:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/949717/2/20250515060855/NEC-V-FC : SUCCESS in 4h 33m 12s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/949717/2/20250515060855/NEC-V-ISCSI : SUCCESS in 4h 49m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f8c8f0553723cf5bd6b1b3919dd6b10853cbfb6f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-16 02:23:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/949717/2/20250516051148/NEC-ISCSI : SUCCESS in 5h 43m 22s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/949717/2/20250516051148/NEC-FC : SUCCESS in 6h 11m 51s","accounts_in_message":[],"_revision_number":2},{"id":"dbcaa46089549ac7b81a98234429b6d1750b35b9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-27 11:49:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5db04e9ceb242db7bf4512741d714f5282736cf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-27 11:59:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cebebc3ec1dc7954ed51776e289556623d5ec2b9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-27 12:02:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6409b1489f6bc99f5520fb911caf71a7ffe96ba6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 12:34:54.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/322/","accounts_in_message":[],"_revision_number":3},{"id":"700c76ab6517120b6da2638640986282fcb469c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 13:30:03.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949717/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949717/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"b3cac648b8400edc8b9bccf38355a70d0f1b3c8b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-27 13:41:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1584/.","accounts_in_message":[],"_revision_number":3},{"id":"1dd99ac3e3f54e32626f89cde6fdaca267fd1544","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-27 13:47:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1719/.","accounts_in_message":[],"_revision_number":3},{"id":"fb143216abcf5ef00ef0f2e46512878b43f8c1b4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-27 13:47:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9caa85f72a4b41abb80f6c2cc78e21c7 : SUCCESS in 47m 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3edb450690f744e6b83e683dfbe00be3 : FAILURE in 59m 46s","accounts_in_message":[],"_revision_number":3},{"id":"02005b83a8e80df9045cc5b89301540179784d00","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-27 13:49:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949717 : FAILURE in 37m 59s","accounts_in_message":[],"_revision_number":3},{"id":"fb17bddbdc03817249ce037fe65a66836dc4a97a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-27 13:52:52.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"06c2f0ee35445fcf83e9c0e67c1f0830a097e42b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 14:52:26.000000000","message":"Patch Set 3: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0616e71c9b44e56b9e09e3a53fe0a99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f7ab8f035504fe28c5b3f2452d9a565 : FAILURE in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8f2f58139ab407e903e2f3150c72d79 : FAILURE in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/161ddf21ebf74f33a7d241a862638207 : FAILURE in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a377e55ebdaf4cb282a081ce69903103 : FAILURE in 9m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d17f2ded692419a90c53278b81c4648 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/f1d49448927340e9b6410f2f821daaa7 : SUCCESS in 34m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ac51fc3023d24c78b90bc85d3af673fe : SUCCESS in 1h 05m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b82c459812634866a417e989e3efd915 : SUCCESS in 1h 22m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6349378353464ba0b14d76ecf1819769 : SUCCESS in 52m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1bfc4a4fa91a4e4eae93888dc591dbd2 : FAILURE in 6m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0ddcc0dab8104ad9aa71b9cfb3a93b56 : FAILURE in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bbd3a731c9b14d80b7b89d31ccd80f36 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a540d152834a44519e84213b76642884 : SUCCESS in 7m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bfe43ec408b948bf8364d364c98e4142 : SUCCESS in 8m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/989711a9e5184b61a16005db209dfd41 : FAILURE in 36m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f36df00141294f749603b15459f59105 : SUCCESS in 7m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ee9f6943a50946dcaf6a4349627974ac : SUCCESS in 2h 01m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/75cff590c8654aa39cce40d8f75117be : FAILURE in 2h 47m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f25755b650264ba4bf9e4e8107e17001 : SUCCESS in 1h 52m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2021d2225b7c43aeb584af1d454ffe3f : FAILURE in 8m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eec7c4633dda412f84e09eaab486d36c : SUCCESS in 27m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f3ddadcc40874de49154852db0ee2b06 : SUCCESS in 1h 15m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/622fbf32fd6d4f00a52ec00adfe5a152 : SUCCESS in 1h 57m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/abd7b345d42641a2833cc987c045458a : SUCCESS in 25m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7095d5f36e854b12bcb57dd7ee2e2078 : SUCCESS in 1h 30m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e453f99352844bddb6b666d3c611d093 : FAILURE in 17m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/63d5321a619047d9a6e28697845900e1 : SUCCESS in 1h 40m 56s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/806e67c1568742c993ff92657d3a9c1b : SUCCESS in 1h 07m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbac790c91a3496a8e4c1623e337eb6c : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":3},{"id":"3e598d7dd358bb62e81a7793794a9633a40d994e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 15:22:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52114d1a8d274e4e8a8c5df35564e981\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f3a04c88d5e4d33ac523f80b487993f : FAILURE in 10m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3a6f8401653b49c7aef22826f733f8ca : FAILURE in 14m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a817a1aa1f6a42b1965640800b5714de : FAILURE in 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8195c23862b6ea2140e1c0e9ba9fe75295d46796","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-27 16:53:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"dc52cb1f8b8f442b705745a10fc889f5e288fb42","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-27 17:45:32.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-949717-3/FC/ : FAILURE in 5h 44m 4s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-949717-3/iSCSI/ : FAILURE in 4h 40s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"356b996f0117e340cfc4bdca047078907c370c11","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 18:32:01.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/326/","accounts_in_message":[],"_revision_number":4},{"id":"54380a216ac687bf5699b34597cc2664df57ac49","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-27 18:42:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/24998d46a63d4f3c94143af2db09c892 : SUCCESS in 48m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/774ae79474c74028813d693399c1f10c : FAILURE in 51m 35s","accounts_in_message":[],"_revision_number":4},{"id":"cd05b283ca2b723ee935155765cab2da457abb70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 18:53:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/245f5d906b55422ebeb4c4a98aa17159\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e4d01a0a34a64e808a7b632de1d1188d : SUCCESS in 26m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9bd4b81ec114413cb3203c7b905fe0d7 : SUCCESS in 26m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f4dac62651fa4b649b4995c85a6a7358 : SUCCESS in 37m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e1b60bc671d7fe0a51813350a47b798793f7e182","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-27 19:02:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949717 : FAILURE in 38m 23s","accounts_in_message":[],"_revision_number":4},{"id":"ec24ee9103d50224eaa5a9a019a7a45252583d89","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 19:26:19.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949717/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949717/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"5d06a60577abcca22fe5a59d4f9d50929f49e14d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 19:43:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e65bc58e1924791826c6bc7b9cb2fda\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75b96c25d0bc4b93960a6b0d7abf4a15 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3912487457614c50a453dd2cf91d75e9 : SUCCESS in 11m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ddda00198f04422921bac9b12146fa2 : SUCCESS in 13m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d86efe9402c548fdafce4666a38645ff : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a889345ea6944b08bad343f8dd4fd21 : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/d50f2a5323734948bf46722e0d071be1 : SUCCESS in 1h 06m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ca5647791de4065928d4163ffb6357e : SUCCESS in 57m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ce468d5ad89400ca9fe7b6912729cf2 : SUCCESS in 1h 23m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed323255d7434caca08c33e28ecafae2 : FAILURE in 21m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/216a5048ea8b4bab9c9c56e01c886516 : SUCCESS in 26m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/50d3504170fb48b6b23c147c42c74ed4 : FAILURE in 7m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2a1a31b8deae4a97abbfdd99bfd0df77 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/889e93ef3e7344669f4cf3f7bf363635 : SUCCESS in 6m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/68f5fdbe823141489c1dd6adfb265ca7 : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/65eff55ea749466cb190769ff16a2853 : FAILURE in 37m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d8728ac2c7d8469d937b0db8e7dcfde6 : SUCCESS in 4m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a0da5c75b2e14d51bc0e88b137b68d99 : SUCCESS in 1h 32m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a103fb5815d84116bc6fc45f660af89d : FAILURE in 2h 43m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8afd91e1ccfa440d8e1528d66f9f2e1f : SUCCESS in 1h 35m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/22702072a6c64ef0af2dc0f16e9b529a : FAILURE in 11m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f153bf7efbf243d29635fbc5fca61bb4 : SUCCESS in 32m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/df3770e4886f415a9b1f1991af0dd060 : SUCCESS in 35m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e4ca5635bb0248cba3c56e7bc0855231 : SUCCESS in 1h 43m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c5eb42c487794e2e92769cda61695fd5 : SUCCESS in 38m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b1709ee3421a427bbca21544abf83024 : SUCCESS in 1h 24m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/05128fa05d3c46b5857c7ec17c86a183 : FAILURE in 18m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/709045ffda104c058be653d16c325221 : SUCCESS in 1h 52m 31s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ad9256893b8438ba012c5a4f58a6651 : SUCCESS in 1h 27m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e1ec769206f4ea2858dc4a28fc748dc : SUCCESS in 51m 10s","accounts_in_message":[],"_revision_number":4},{"id":"2e1e2c973084f922c458a9c1e01941d8e5a96dcd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-27 20:22:12.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"242cbccc0a74636e0ca15dc0a56b00b533fc2337","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-27 21:00:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1588/.","accounts_in_message":[],"_revision_number":4},{"id":"0d82d11789c7e53d4bcdd7434e2fa58279fe642e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-27 21:07:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1723/.","accounts_in_message":[],"_revision_number":4},{"id":"4079f7f5d4ff3fc86c251a45152bd991add5fcdc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-27 21:55:32.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-949717-4/FC/ : FAILURE in 5h 47s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-949717-4/iSCSI/ : FAILURE in 4h 7m 26s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"49c3e314599b9e28a4b09cb28b6d21bf05dfd3c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-28 09:30:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"975d4593ff1f3f0f6b9ad17d242f536875f20513","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 09:30:44.000000000","message":"Patch Set 5:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/336/","accounts_in_message":[],"_revision_number":5},{"id":"9e491c0d15e854f46c9f4363265839ecb6acae7b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-28 10:10:14.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/949717 : FAILURE in 37m 55s","accounts_in_message":[],"_revision_number":5},{"id":"d31b3609ff5c10d5d7e60b6621674f2dd54e582b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 10:23:28.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949717/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/949717/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"1bbbf910e44075841d7f78d409f4aba2d9466483","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-28 10:29:33.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e8d6ce4c62474ab3a1ed6f890b8e558d : SUCCESS in 44m 18s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/681bfccca60344b4bca4a9285fe7c994 : FAILURE in 57m 16s","accounts_in_message":[],"_revision_number":5},{"id":"a6c49fbf8546a487f3bbd63f4d5d05d6ce0c8e54","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 10:30:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1596/.","accounts_in_message":[],"_revision_number":5},{"id":"8b537bce8024c908524eb59adf57a401cf47b22e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 10:33:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1731/.","accounts_in_message":[],"_revision_number":5},{"id":"68299e9e4146dc3235b97fc7cd9e577cd17fecd1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-28 11:23:02.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"743620df971fafe8fe836c998043407b66d96423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 11:58:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa59561c3502455896a1ee33384c3b37\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a18ba1b045014795bf0453c58541aff8 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5202b2ae3a534a0a8b611fff33f14741 : SUCCESS in 11m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ac9e7415e734d32afc50b9a18058381 : SUCCESS in 11m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be23a095ab2749f3a276dfe43551447d : SUCCESS in 15m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e8a6aa99c47411bb1cfee713eea5c27 : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/9d5b1a30fa33415db5e9dc1fcf0354e4 : SUCCESS in 34m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f64063eac004cbab98e2e957de48030 : SUCCESS in 1h 00m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2db653fedc4a41d4b2c6da0ce1f27cac : SUCCESS in 1h 25m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5e8274d2d384d1ca1a67eae81b9b292 : SUCCESS in 36m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a875994ad6f5499cb8293f22a35d3f7c : SUCCESS in 22m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a1a822cdc3c74b51b5ed22305f378c91 : SUCCESS in 7m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bb27ec97877244ad8d930b8dca6ac204 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/30d1a7827c3b4cccac9f85632d57bcbc : SUCCESS in 5m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dd259a41cc8f453293957426f40d7068 : SUCCESS in 10m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9d4e853a60d4112b60a262e445b0998 : FAILURE in 35m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/84045f946b2b492fad56d3f5a59ad7e3 : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e8689c470999426eb863a3cdf4c9f622 : SUCCESS in 1h 35m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2148cafb51ff405f8c7a5b1c5d14660c : FAILURE in 2h 25m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/52746a8e26a14a1d8b3ab92bd2f8829c : SUCCESS in 1h 40m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7972f3f15fa542f1af92cc341db1dd10 : FAILURE in 17m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cab93c2d105e48b0a45dec1bf38b5b12 : SUCCESS in 33m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dac9043f9d9344b291b3d72fcf23f622 : SUCCESS in 1h 13m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4774603283234e2999821c971e5a5086 : SUCCESS in 1h 23m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dba134ef17d24cf982a1c275c9307707 : SUCCESS in 39m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/458945574d23477c84dbbe8705519de8 : SUCCESS in 1h 30m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a5c48d64c75b48cc86eb62dcf62bd6b6 : FAILURE in 19m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a2865ab38a6d402e85bb79aa861f6bac : SUCCESS in 2h 05m 43s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd4f6a3743664345bb113f8b1792949a : SUCCESS in 1h 34m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5dc751e335984ad0acd6785de4864a15 : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":5},{"id":"f11aa3505a10904bda6afad2c8a14a8875521220","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 12:20:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2127411d9695422197060daa522477b4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca544930fd7c40d59c3e3c011d49c438 : SUCCESS in 24m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f8d168b4c081445ba28772dc4780d6a9 : SUCCESS in 29m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/91d0068d8cbd4a43ac2d73ec7ff25017 : SUCCESS in 35m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5bba192dba957ebf63caba887820001561629595","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-28 14:37:24.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-949717-5/FC/ : FAILURE in 5h 3m 53s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-949717-5/iSCSI/ : FAILURE in 3h 52m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"ee1c9847233cc688dc367f7dcfbbe62150a3a1c5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-09-05 01:55:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/949717/5/20250905011722/NEC-ISCSI : SUCCESS in 5h 47m 30s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/17/949717/5/20250905011722/NEC-FC : SUCCESS in 9h 37m 39s","accounts_in_message":[],"_revision_number":5},{"id":"c4b2d925b47dc023b3e2de23fb7850aa76e7629a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-25 09:06:39.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"54a2bbf5da429be3538ca944856c301b00ee2ea3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-25 09:41:50.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/f881a7622f3d4873b42f237882fd905d/","accounts_in_message":[],"_revision_number":6},{"id":"b3906ec848edb4606c1260bef22459add3d5529c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 09:57:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28184493f2954778b1ce45ba71adc119\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1c95ba487074723b5d2f9707b7ddc30 : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b0635f2acfd542458da6e3cb09f74e1f : SUCCESS in 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d61cd91ecc032e2b7f80916de475e4671ee98665","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-25 09:57:45.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f881a7622f3d4873b42f237882fd905d\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ee79fa3220cb4c8b8f7db90eb151833f : SUCCESS in 37m 24s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8c1c5d38166942bc90723895fd6105fa : SUCCESS in 38m 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6b91b3465b9d4110855742641deaeb21 : SUCCESS in 47m 45s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ef9d531bf07f4c3baed4c5f8f3261df2 : SUCCESS in 31m 28s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9fbc8f093fa84b1abca991b593908f56 : FAILURE in 15m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0e3226c5f8852bcb62bae8b30999761bad95c29f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-25 10:06:15.000000000","message":"Patch Set 6:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 1m 46s\n- FC       : FAILURE in 0h 1m 21s\n- NVMe/TCP : FAILURE in 0h 1m 27s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/949717","accounts_in_message":[],"_revision_number":6},{"id":"a5365dd4204362b96823183e2a292bf59b82c6c1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-25 10:53:10.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/e1157110a9db49bb94792b151efbd282 : SUCCESS in 47m 26s\n- cinder-storpool-tempest-iscsi-multipath build/95254e084cbd4a5e8ac7dea82a71bd0f : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":6},{"id":"d869d04b5ec598814450c06960ccdb0d9948c575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:40:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b47c66ed96a24e439a86d5e0554c8547\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64941c86427540588d0196e6ab8cc61c : SUCCESS in 8m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69351828936e47f693c2cdb150fd546e : SUCCESS in 8m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c0d2caeeb464b7fa07408454b64ff31 : SUCCESS in 14m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0955a3a4c204b4cb7c7ff46fe97f248 : SUCCESS in 15m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/cedb9005b81e4edfb628eaafb367cd32 : SUCCESS in 1h 05m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/06412d0dbb7d4935bf72237db48c2285 : SUCCESS in 37m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1f799f23ce314cbaa3bacc28d7b2237a : SUCCESS in 57m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ba295b10ba9486e8ef0fb6682a6bb4a : SUCCESS in 1h 07m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ba80eebfc8549b992bbc389612a1268 : SUCCESS in 6m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/687a382ec6074ccaad6315311535bd5e : TIMED_OUT in 41m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b08737c4dce24335b69e7727ee688d2c : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a1ca05618274ff8b828f35412bf4100 : SUCCESS in 6m 59s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/b2b2da32534447cbba76d450345ea440 : SUCCESS in 9m 40s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/704de1f451cb44d2b1e9daeed006b2a6 : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/df6bd12e3abf45acad765a680dbeda9b : SUCCESS in 6m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8dab874ae23c4099b2d0b4ca129078a5 : SUCCESS in 6m 39s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/dbd1d518837547439781f2d8b4c41dd6 : FAILURE in 5m 30s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b5276da988c4430793272a4601d3ece2 : SUCCESS in 9m 34s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5a6554b555ad4f50922388f1f9d95675 : SUCCESS in 19m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ef4748e4c9004d33b18029e9a51b81c3 : FAILURE in 3m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/be9d217f98114de292c0ef00d8c3e4f6 : SUCCESS in 1h 37m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1e305c7d68bc43d18646e434f5fd6e0b : SUCCESS in 1h 11m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3e56e3baa21d411ba6eb7143a07c742c : SUCCESS in 1h 50m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/85640c2a618143138a7368781a1d3eec : FAILURE in 2h 26m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/09f696aa79a84a2088d31d2713fdf910 : SUCCESS in 1h 22m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/79334281ce4740c299ee5ed4905526a3 : SUCCESS in 31m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bda7ac8cd9074d3b8c737620829d1cac : SUCCESS in 1h 19m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2e7f79e39cd84f628dcab0736d8537bd : SUCCESS in 1h 30m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/48e3be0b63f0470dacd4df941338dab9 : SUCCESS in 54m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c67d75f8db744f7b8c6680d16b887147 : SUCCESS in 2h 16m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0cf2d266b23a47be97a7138a8d50a6ef : SUCCESS in 1h 47m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b6febaec73a44769b6b97651a19e7e5 : SUCCESS in 1h 01m 25s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"e2fef24e6dfb929ef8adae9ef0de8823963bf424","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-25 11:49:46.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_09_2026_11_43_53\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"4afdd4548f6f0172850239733d4a4677842ea3bf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-25 14:30:54.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e2551f087b589c3e3e094f70169360d74099e57 .","accounts_in_message":[],"_revision_number":6},{"id":"6b6265f61f4a767484e9456e7de7dbf01c30ce20","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 02:26:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5708/.","accounts_in_message":[],"_revision_number":6},{"id":"e513e68fb2d9859f8a459cc74080f92016cab171","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 03:26:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5531/.","accounts_in_message":[],"_revision_number":6},{"id":"d328d7a5c6ba2f66b13eb64e862ad2c1f6e5ae8f","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-27 17:10:51.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c6b1504ce1da04f0abe456d9f686a1237d1ae388","revisions":{"62eb82b398c0dfa8237f2b749fa4fd6386e31236":{"kind":"REWORK","_number":1,"created":"2025-05-14 09:56:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/949717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/949717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/949717/1"}}},"commit":{"parents":[{"commit":"838cb6211af0e2447972036c3d87c4a1697a6ab4","subject":"Do not convert images when format is the same","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/838cb6211af0e2447972036c3d87c4a1697a6ab4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-14 09:55:54.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-14 09:55:54.000000000","tz":120},"subject":"Avoid deep copying of volume when it\u0027s on the same filesystem","message":"Avoid deep copying of volume when it\u0027s on the same filesystem\n\nWhen both source and destination of the image-\u003evolume are on\nthe same filesystem, like in case of image conversion, and does have\nsame volume type, it makes sense to perform a move of the image\nrather then deep copy using dd.\n\nIt not only speed up the download process, but also reduce diskspace\nconsumption during such migration.\n\nA sample example would be having images on Swift storage while\nusing Ceph as a volume storage. In case image is already stored in RAW\nit is enough to fetch, verify and move the image to expected destination\nrather then also doing deep copy with dd.\n\nChange-Id: I6e2551f087b589c3e3e094f70169360d74099e57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/62eb82b398c0dfa8237f2b749fa4fd6386e31236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/62eb82b398c0dfa8237f2b749fa4fd6386e31236"}]},"branch":"refs/heads/master"},"50a2f0e9dd8c88bcc09ccca55a97558bbe032102":{"kind":"REWORK","_number":2,"created":"2025-05-14 09:58:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/949717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/949717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/949717/2"}}},"commit":{"parents":[{"commit":"838cb6211af0e2447972036c3d87c4a1697a6ab4","subject":"Do not convert images when format is the same","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/838cb6211af0e2447972036c3d87c4a1697a6ab4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-14 09:55:54.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-14 09:57:30.000000000","tz":120},"subject":"Avoid deep copying of volume when it\u0027s on the same filesystem","message":"Avoid deep copying of volume when it\u0027s on the same filesystem\n\nWhen both source and destination of the image-\u003evolume are on\nthe same filesystem, like in case of image conversion, and does have\nsame volume type, it makes sense to perform a move of the image\nrather then deep copy using dd.\n\nIt not only speed up the download process, but also reduce diskspace\nconsumption during such migration.\n\nA sample example would be having images on Swift storage while\nusing Ceph as a volume storage. In case image is already stored in RAW\nit is enough to fetch, verify and move the image to expected destination\nrather then also doing deep copy with dd.\n\nAt the moment 50Gb RAW image will consume 100Gb diskspace on\ncinder-volume for conversion, which is not even required.\n\nChange-Id: I6e2551f087b589c3e3e094f70169360d74099e57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50a2f0e9dd8c88bcc09ccca55a97558bbe032102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50a2f0e9dd8c88bcc09ccca55a97558bbe032102"}]},"branch":"refs/heads/master"},"39323812c9368effd0e6850f6e653c91f6568f8f":{"kind":"REWORK","_number":3,"created":"2025-08-27 11:59:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/949717/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/949717/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/949717/3"}}},"commit":{"parents":[{"commit":"e23867d17e858d5fc6f6042d0d410111089b3ba8","subject":"Do not convert images when format is the same","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23867d17e858d5fc6f6042d0d410111089b3ba8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-14 09:55:54.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-08-27 11:59:25.000000000","tz":120},"subject":"Avoid deep copying of volume when it\u0027s on the same filesystem","message":"Avoid deep copying of volume when it\u0027s on the same filesystem\n\nWhen both source and destination of the image-\u003evolume are on\nthe same filesystem, like in case of image conversion, and does have\nsame volume type, it makes sense to perform a move of the image\nrather then deep copy using dd.\n\nIt not only speed up the download process, but also reduce diskspace\nconsumption during such migration.\n\nA sample example would be having images on Swift storage while\nusing Ceph as a volume storage. In case image is already stored in RAW\nit is enough to fetch, verify and move the image to expected destination\nrather then also doing deep copy with dd.\n\nAt the moment 50Gb RAW image will consume 100Gb diskspace on\ncinder-volume for conversion, which is not even required.\n\nChange-Id: I6e2551f087b589c3e3e094f70169360d74099e57\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39323812c9368effd0e6850f6e653c91f6568f8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39323812c9368effd0e6850f6e653c91f6568f8f"}]},"branch":"refs/heads/master"},"68f31fddd44d2e1bbed071658afb9637c8c41901":{"kind":"REWORK","_number":4,"created":"2025-08-27 16:53:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/949717/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/949717/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/949717/4"}}},"commit":{"parents":[{"commit":"e23867d17e858d5fc6f6042d0d410111089b3ba8","subject":"Do not convert images when format is the same","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23867d17e858d5fc6f6042d0d410111089b3ba8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-14 09:55:54.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-08-27 16:53:32.000000000","tz":120},"subject":"Avoid deep copying of volume when it\u0027s on the same filesystem","message":"Avoid deep copying of volume when it\u0027s on the same filesystem\n\nWhen both source and destination of the image-\u003evolume are on\nthe same filesystem, like in case of image conversion, and does have\nsame volume type, it makes sense to perform a move of the image\nrather then deep copy using dd.\n\nIt not only speed up the download process, but also reduce diskspace\nconsumption during such migration.\n\nA sample example would be having images on Swift storage while\nusing Ceph as a volume storage. In case image is already stored in RAW\nit is enough to fetch, verify and move the image to expected destination\nrather then also doing deep copy with dd.\n\nAt the moment 50Gb RAW image will consume 100Gb diskspace on\ncinder-volume for conversion, which is not even required.\n\nChange-Id: I6e2551f087b589c3e3e094f70169360d74099e57\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68f31fddd44d2e1bbed071658afb9637c8c41901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68f31fddd44d2e1bbed071658afb9637c8c41901"}]},"branch":"refs/heads/master"},"f565a5cd03ea65e63ec772b9ea824be429e12464":{"kind":"REWORK","_number":5,"created":"2025-08-28 09:30:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/949717/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/949717/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/949717/5"}}},"commit":{"parents":[{"commit":"e23867d17e858d5fc6f6042d0d410111089b3ba8","subject":"Do not convert images when format is the same","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23867d17e858d5fc6f6042d0d410111089b3ba8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-14 09:55:54.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-08-28 09:30:23.000000000","tz":120},"subject":"Avoid deep copying of volume when it\u0027s on the same filesystem","message":"Avoid deep copying of volume when it\u0027s on the same filesystem\n\nWhen both source and destination of the image-\u003evolume are on\nthe same filesystem, like in case of image conversion, and does have\nsame volume type, it makes sense to perform a move of the image\nrather then deep copy using dd.\n\nIt not only speed up the download process, but also reduce diskspace\nconsumption during such migration.\n\nA sample example would be having images on Swift storage while\nusing Ceph as a volume storage. In case image is already stored in RAW\nit is enough to fetch, verify and move the image to expected destination\nrather then also doing deep copy with dd.\n\nAt the moment 50Gb RAW image will consume 100Gb diskspace on\ncinder-volume for conversion, which is not even required.\n\nChange-Id: I6e2551f087b589c3e3e094f70169360d74099e57\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f565a5cd03ea65e63ec772b9ea824be429e12464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f565a5cd03ea65e63ec772b9ea824be429e12464"}]},"branch":"refs/heads/master"},"c6b1504ce1da04f0abe456d9f686a1237d1ae388":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-09-25 09:06:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/949717/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/949717/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/949717/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/949717/6"}}},"commit":{"parents":[{"commit":"9adf98f6ff33a77b0eadb82d42a4aa28ccb67cd9","subject":"Do not convert images when format is the same","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9adf98f6ff33a77b0eadb82d42a4aa28ccb67cd9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-14 09:55:54.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-25 09:06:39.000000000","tz":0},"subject":"Avoid deep copying of volume when it\u0027s on the same filesystem","message":"Avoid deep copying of volume when it\u0027s on the same filesystem\n\nWhen both source and destination of the image-\u003evolume are on\nthe same filesystem, like in case of image conversion, and does have\nsame volume type, it makes sense to perform a move of the image\nrather then deep copy using dd.\n\nIt not only speed up the download process, but also reduce diskspace\nconsumption during such migration.\n\nA sample example would be having images on Swift storage while\nusing Ceph as a volume storage. In case image is already stored in RAW\nit is enough to fetch, verify and move the image to expected destination\nrather then also doing deep copy with dd.\n\nAt the moment 50Gb RAW image will consume 100Gb diskspace on\ncinder-volume for conversion, which is not even required.\n\nChange-Id: I6e2551f087b589c3e3e094f70169360d74099e57\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b1504ce1da04f0abe456d9f686a1237d1ae388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b1504ce1da04f0abe456d9f686a1237d1ae388"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e23867d17e858d5fc6f6042d0d410111089b3ba8","ours":"f565a5cd03ea65e63ec772b9ea824be429e12464","theirs":"9adf98f6ff33a77b0eadb82d42a4aa28ccb67cd9","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
