)]}'
{"id":"openstack%2Fcinder~950927","triplet_id":"openstack%2Fcinder~master~Ibf260fe254b048d9aeb038c388018b26293f8d8a","project":"openstack/cinder","branch":"master","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-10-10 11:57:40.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2026-05-29 00:07:17.000000000","reason":"Someone else replied on the change"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-05-28 23:24:31.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-05-28 23:24:31.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ibf260fe254b048d9aeb038c388018b26293f8d8a","subject":"cinder-manage: Perform database purge operations in batches","status":"NEW","created":"2025-05-27 02:12:25.000000000","updated":"2026-06-19 19:49:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":13,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"f643e10e6de1791f81307cc08ebde7e603ee1735","_number":950927,"virtual_id_number":950927,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-29 02:11:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-27 02:49:11.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-05-27 02:55:36.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-05-27 03:04:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-27 03:06:48.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-27 03:10:24.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-05-27 03:41:12.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-05-27 03:44:51.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-27 04:04:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-27 05:21:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-27 21:14:30.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-05-28 00:32:52.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-05-28 05:42:47.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"CC"},{"updated":"2025-05-29 03:17:55.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-06 22:07:23.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-06-07 03:08:13.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-06-07 08:07:04.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-06-07 13:07:37.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-06-11 13:29:30.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-07-16 20:40:36.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-10-10 15:36:46.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-11-05 14:27:27.000000000","updated_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"reviewer":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"state":"CC"},{"updated":"2026-01-30 20:11:55.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-01-30 22:44:08.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-02-03 00:24:50.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-04-08 15:33:01.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-05-29 00:07:17.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-29 00:10:59.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-05-29 04:18:10.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-06-19 19:49:41.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"10ddd3e4c6a908a61c171dff1a60d201d44b615e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-05-27 02:12:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64a2f500bb7329782362da2339b3aeb847ab09a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-05-27 02:13:56.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e967854ae5201285666f4cafa839cfac8a3f476f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-27 02:49:11.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/761d8a788be240899a9417a4389e6b1c\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a5e234edda264664805bdf92cdd9df9f : NODE_FAILURE Node request 200-0000054606 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6b116c2c7487458a91af8e5020681ce3 : NODE_FAILURE Node request 200-0000054607 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6b6f1adb6adc474189f47e79d0b22c1c : FAILURE in 26m 44s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/909c3725609c4b3aa2b395c17e8731d3 : FAILURE in 27m 19s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9e6c451b4d9b4ce08d4faf7409c168e4 : FAILURE in 28m 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/379b957f9c5f4d5e89a36f20b335ddad : FAILURE in 25m 39s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/be0e944ab52f45349dbfb760d9a4aaf6 : RETRY_LIMIT in 34s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7d7ea9b500904d2089016d6e23255bd3 : RETRY_LIMIT in 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b23245d2e131462084501c7fc2b644a9 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6636a5cbc7844b1cae29747a876e2c36 : RETRY_LIMIT in 28s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source... (truncated)","accounts_in_message":[],"_revision_number":2},{"id":"e37bc1726e617f7a13032cd76b842eeb9dc11f40","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-27 02:55:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950927 : FAILURE in 37m 23s","accounts_in_message":[],"_revision_number":2},{"id":"602a3c155ca0a04f8ddb5cb604c44b613baab3b0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-27 03:04:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950927 : FAILURE in 44m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6abe176ca5274257df29cee724b8418893a3647d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-27 03:06:48.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c564d7c3e76f183826ed8597edbaf52ea1944f2e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-27 03:10:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/915/.","accounts_in_message":[],"_revision_number":2},{"id":"d56657e20d07ba73da834143a03023038210b26a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-27 03:20:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/789/.","accounts_in_message":[],"_revision_number":2},{"id":"4263543c9612292ae302f9f133b9d8c042ac19fb","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-05-27 03:41:12.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-27-950927-1/FC/ : UNSTABLE in 24s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-27-950927-1/iSCSI/ : FAILURE in 1h 28m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"7c2e70ae83cf67fc77591bac68184054e2af5d5b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-27 03:44:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6e7747720ad94d839216ca676b039353 : SUCCESS in 1h 25m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/80fbd4e68e5c4022bfd0e7c36362ccb8 : SUCCESS in 1h 29m 57s","accounts_in_message":[],"_revision_number":2},{"id":"24711100f938a5e315b27116901fcdfc3475c330","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-27 04:01:02.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"bb1e2f7b3505fd462354955f5147c73100299e75","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-05-27 04:02:20.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-27-950927-2/FC/ : FAILURE in 1h 30m 53s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-27-950927-2/iSCSI/ : FAILURE in 1h 48m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"93d8502a242dfb938b28daad6b91a161ba948189","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 04:04:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55b7772c254544b9975665a8517edd8d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b999e73bdc1c4fddae4636b6bf1fd4c6 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a3970f17c59d45d4801ad9a36f0b5025 : SUCCESS in 26m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2b0b2a5f52eb4a748bef77c17be515ef : SUCCESS in 14m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ac60483a06f62b275161005df37eac3f152de08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 05:21:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a99f31ed65e349b189126a1fa05fab95\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/895a0bc945464e3bb189784f205bd223 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6167654746b4af6a4a25ff9af032b5c : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22c4f4835f2a488e80808f546684e844 : SUCCESS in 10m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ceab549ef13f45fc8c8e351cfcc212e4 : SUCCESS in 12m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3e28a07d777477da2dfd61520acf1aa : SUCCESS in 14m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a917996bb001442c80b09d3e2cae3bd0 : SUCCESS in 13m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/9d30077aa07d411b8cd2eaff5d9d210f : SUCCESS in 52m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/54462196149c4d6ea87f6327b9dc9253 : TIMED_OUT in 3h 02m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4079b69a67b4bbcbcce1a3b489d4c10 : SUCCESS in 1h 37m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ad2efd7f2b04d3790790a7f2d5aa6c8 : SUCCESS in 58m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23668efe78024f6caf10e4d8f58a8c63 : FAILURE in 3m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d787d885ef1e4313a7052dadca50dd61 : SUCCESS in 28m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5d52152c275f48318573ed1a884e47c7 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5edade4c0848449883124b2ccd87a94d : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f877e29a66b2404494e50fa3a66c49af : SUCCESS in 7m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3a6ea7ecfa424db89252fa191206464b : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cfb4028d74ef44f6a3155d2a91b2eb74 : FAILURE in 33m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62230e000b1e4ed38433e6f86de7b908 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e2f9a03bd9df4969b54439db6e109fdd : SUCCESS in 1h 46m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d069214f6da94333a9f2082cc5e70832 : FAILURE in 2h 10m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/89ccb23a1e64450697a94baae0f86e65 : SUCCESS in 1h 41m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6b7fed3fc269420f86a32e3d9fe93d44 : FAILURE in 26m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f858394c9c0145f28a6d3ce336c330f9 : SUCCESS in 31m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d5185cead85f46cb96dfd1e6428395fd : SUCCESS in 55m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8214e343cb67440bbbebf96fd260d3c0 : SUCCESS in 1h 35m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/904faf1da0a140e5abec68826160b921 : SUCCESS in 48m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/97151d9c72c64f1da5e83d04dd43b2b8 : SUCCESS in 1h 32m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/90351720037e46178626ea6b9fa733cf : FAILURE in 20m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8ed48b9169a4617bd24b480ddde6a76 : SUCCESS in 1h 41m 48s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18a27e284b524e4f9a0a1802adc6ab7b : SUCCESS in 1h 26m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b823c0e6e91438bb48c6271b17af286 : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a8989ff919da4d8e0a7f8937eff21283586efaf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-05-27 15:53:55.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"67342def36b9c4f65de8d084219ad52c8acc50ee","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-27 16:10:05.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/921/.","accounts_in_message":[],"_revision_number":3},{"id":"22a9a136d35b25e05ac9e76c4900070e4b2653d7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-27 16:10:22.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/795/.","accounts_in_message":[],"_revision_number":3},{"id":"02ff171a3bb23d5054b4d35bc93179d26fbe274e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 16:18:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cd83b2ea3e74d0f8d7d4be02ae78e61\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f7bf631757bd464f9d203f81e26df131 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4b4c8dbca7ee41f5b28d2422ba006bf7 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a8fec14959d94c9398712a604757ca4d : SUCCESS in 22m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d8ee48d0d4dc9eb2c07be48402662ea8923ebb65","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-27 16:26:55.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/eb7ad4dcaeef4b378ef655921f61f9cc\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/55bc7fd63b65480787aa6e881d325d92 : NODE_FAILURE Node request 200-0000054691 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d3b687490bbf460f90a206d9ebb63aa5 : NODE_FAILURE Node request 200-0000054692 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7a2e6a93776e4b1e90ebec0f155f147c : FAILURE in 26m 27s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/adbaafba96d74011ae44989e1f136803 : FAILURE in 23m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/327c4b4268424f8486d39b7b3e6cd94e : FAILURE in 25m 17s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b36ea1a6651b48b3ab6f711d567e0475 : FAILURE in 24m 39s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/efaa92dede124c24a825aaa7c6addd65 : FAILURE in 31m 26s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4d327688bd5a4cbdb97c6bf9cec509a8 : FAILURE in 32m 10s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9ff5555bc942408594d47478dc093070 : FAILURE in 31m 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/31a3f238894c429db43ec21bd8f8d97a : FAILURE in 32m 24s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source... (truncated)","accounts_in_message":[],"_revision_number":3},{"id":"45a610d22ebae38a7a0e360487a7aa0a122d94e5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-27 16:42:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950927 : SUCCESS in 47m 00s","accounts_in_message":[],"_revision_number":3},{"id":"7d3a6f4fb7ee3574bc8751f661511a2afa1a6814","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-27 17:13:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950927 : FAILURE in 1h 17m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a30e77c53fc42c368ec74d5985209dd6372603f0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-27 17:18:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9babaca16c3349a1a7522f3194961776 : SUCCESS in 1h 22m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6ce7b7ab8f4440b085b93655324afe31 : SUCCESS in 1h 15m 51s","accounts_in_message":[],"_revision_number":3},{"id":"f68346cf5cd503293521392035a338265c972f08","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-27 17:38:23.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"0cc68df0e4dee4fce19e9209fb2167d320ed0ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 18:12:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9385c6626cf74f0d97350e4a8c00cbaf\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9f42fc5e48664965b87b3bff6719b40a : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/326032710a6a46e58da4eb7b9980347e : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5e90ffcf3f44d9aae7ce76f2fd2ed9e : SUCCESS in 11m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88fa4e16084740aa81e381d8a4d4569f : SUCCESS in 11m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97af6e8677834c8f99e4da196261e25f : SUCCESS in 15m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2610d3b03ba347af88e6cc22ef2f48aa : SUCCESS in 14m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/bb91c4a7ad44418fa29288e2dcfe606d : SUCCESS in 44m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21f2dc2df6b14e47858c347a0da39491 : SUCCESS in 1h 02m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/286aafc4f5fa42bf8e32b6df24c0769f : SUCCESS in 1h 31m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a99a66b69674dc69b52784364a3acc9 : SUCCESS in 52m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc69a0ada3b14f05bce02ebab3af7edf : SUCCESS in 15m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5883174260e34f7b90cb843c42688bf6 : SUCCESS in 24m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/47c9d537274540a18c91c842467bf346 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/53d73d2aecb24db999b2ee5799d44b27 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/46bf0fdcf6c14eabb396136ccf298df6 : SUCCESS in 7m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/11c1085c0c8a4cd4a91665a6b56246b7 : SUCCESS in 8m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2e90a6ffe29344e7a761062c9e6a9514 : FAILURE in 30m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d65fec1571148f699157c3cfbf18588 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/15585616ce4540e2a507c8601c203423 : SUCCESS in 1h 59m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4988cd73cd4f4eda845084180dd615e2 : SUCCESS in 2h 16m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/17d134e4e99b4c529a877e165f030b0a : SUCCESS in 2h 12m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0ace4ced5c464265a3913bc5f66a8257 : FAILURE in 22m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bf4b120a65b94030bfdd4c450be6d8ec : SUCCESS in 27m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a5cc62717ae843058f67019bcdb90ccd : SUCCESS in 1h 07m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/25429b57ab5849da8e8f98937b681cb5 : SUCCESS in 1h 35m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f95c56bf6f094293b6ff3f9486f26a0e : SUCCESS in 44m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/933c3ea8486747c3813013a799e23c5b : SUCCESS in 1h 22m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2769100bac304f12a3d11ef56e2749ae : FAILURE in 16m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a6f8ac2c3a24dc19ed846487839757b : SUCCESS in 1h 24m 33s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c786b18a3fb34a93b553822325cbda8a : SUCCESS in 1h 27m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/683171ea7a9e493bb6790ef58cc84b9d : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":3},{"id":"3321510c8f6746d593e2c558e3f3fd8d2dfbf837","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-05-27 21:14:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"d66ac775d265f7bc3aed83eb5adf478c70c6b8db","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-28 00:32:52.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/2/20250528002656/NEC-V-ISCSI : SUCCESS in 4h 06m 31s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/2/20250528002656/NEC-V-FC : FAILURE in 9h 05m 53s","accounts_in_message":[],"_revision_number":2},{"id":"a7f1a3cb1a9835065a06492bce9eb1edbc4bb216","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-05-28 05:42:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9c7d52a0ff571e993ca53a38fac7606d46c16e58","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-28 06:01:17.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2908237a094fb9ad2b1fead0f5776cee4886546c","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-05-28 08:14:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a08a2cb71a873d26cf1503bc9748bd1da0806f1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-28 08:33:16.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"5da804a482fdc1336bf0fef4b823a77822f255bd","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-28 23:37:42.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/3/20250528224000/NEC-V-ISCSI : FAILURE in 7h 04m 57s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/3/20250528224000/NEC-V-FC : FAILURE in 9h 57m 39s","accounts_in_message":[],"_revision_number":3},{"id":"9e6dad8819a30191e96969cde7fa27ce1c4a4ce4","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-29 03:17:55.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/3/20250529060653/NEC-ISCSI : SUCCESS in 5h 44m 23s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/3/20250529060653/NEC-FC : SUCCESS in 6h 10m 59s","accounts_in_message":[],"_revision_number":3},{"id":"8c45c312a56d0ca6b7b9d1f654080c96c5a277a0","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-06-06 22:07:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/950927-3/Seagate-iSCSI/ : SUCCESS in 2:26:47\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/950927-3/Seagate-FC/ : SUCCESS in 2:11:34","accounts_in_message":[],"_revision_number":3},{"id":"1d87d5eb9be130cd087d7055b4347fa63a9d5190","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-06-07 03:08:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/950927-3/PVME-iSCSI/ : SUCCESS in 2:26:19\n- PVME-FC https://os-logs.seagate.com/dell-ci/950927-3/PVME-FC/ : SUCCESS in 2:12:47","accounts_in_message":[],"_revision_number":3},{"id":"42696f0f501e59175080906f7c639be066c17bdb","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-06-07 08:07:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/950927-3/HPMSA-iSCSI/ : SUCCESS in 2:24:45\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/950927-3/HPMSA-FC/ : SUCCESS in 2:11:56","accounts_in_message":[],"_revision_number":3},{"id":"9b9f6c6ee39548b1a820867a6e67fdf5cb01de79","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-06-07 13:07:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/950927-3/Lenovo-iSCSI/ : FAILURE in 2:25:10\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/950927-3/Lenovo-FC/ : SUCCESS in 2:13:03","accounts_in_message":[],"_revision_number":3},{"id":"c3ff23c39592fa29810ccfd947c261b6bec2a7df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-16 10:15:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"86d68529f6631845382ac21164d5d15035960a32","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-16 10:17:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/16280c4a2d9d42b48c52f2637974530c : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/28ccc90aca5e4eb490b1885aca9abe01 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"9c440fc8b1d6dfd11b1cca9cf03a27203e154358","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 10:34:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6060d5e246bc4eecb3c1ac35337a922a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/476753e4db2f4d7e911a5b6b19b96035 : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/98b00e1caf2c47f696780cea277e20d2 : SUCCESS in 16m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/68b41992336e4f19bb125483a8869dc5 : FAILURE in 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b11e0edd217eee8a84e0e9c0f5d2e2b5a29cde35","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-16 11:16:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/984/.","accounts_in_message":[],"_revision_number":4},{"id":"29a91854face9ecc287f0f4556dd0ee4b43c99b2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-16 11:19:49.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950927 : SUCCESS in 46m 13s","accounts_in_message":[],"_revision_number":4},{"id":"ba8a42ecf411917dd9779c5dcc062cd81e6ec032","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-16 11:19:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1114/.","accounts_in_message":[],"_revision_number":4},{"id":"093cdecb23cc16d02daf3d8c4fa122de6e97d751","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-16 11:47:08.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-27-950927-4/FC/ : UNSTABLE in 1m 51s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-27-950927-4/iSCSI/ : FAILURE in 1h 31m 4s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"ebc2d53387778b310448297e924c097ab904c1a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 12:01:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/243bbb56d8414bccbc50ce8736b8e95d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f26470c01f440a6b575439bc22c2786 : FAILURE in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5affe543bdb4f2cbfcc1c45c950d6bb : SUCCESS in 10m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff2fcf9739d94d94b0ab19d828de5fea : SUCCESS in 11m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4af3bc724a44391856a1ef91b71ca10 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67d5afc8d533496e8629657c713d5625 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/aca16bbfe0384bcf8ae5fc241e9ef5c7 : SUCCESS in 50m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a3d31002c02b4d67ace40f9539786b43 : SUCCESS in 58m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/398ee1b226f0421494ac7742a81801e6 : SUCCESS in 1h 30m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7229f336298a418fb75367f456afd7fa : SUCCESS in 1h 03m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a33288a5efd84f2da5e30f1ab01cce67 : SUCCESS in 15m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/45fb45700723405f8a5855a72237dd6d : SUCCESS in 23m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c9223b32780a439d8f525a7620f06204 : SUCCESS in 7m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f8f389de15584bf48ddc6b0f3a1723b5 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/147f78e029724779a171a9f372ae6e26 : SUCCESS in 5m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/59118c2d676d429eac5c5a066729c5eb : SUCCESS in 10m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/682e0686168e400ca12702d079c0e0f8 : FAILURE in 29m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/27127b300876402eb541f38f201f2bf7 : SUCCESS in 7m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/63bce6ffc38e46a59b2e66398d35844f : SUCCESS in 1h 40m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a35b5bbafe44531b18aa76ccb18d55c : SUCCESS in 1h 44m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e93d2ee316e425780a17af4d414a733 : FAILURE in 1h 43m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fde6540026ca41f68cb213346948e9dd : FAILURE in 21m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/36d2bdfa54e64ddcab953c6b1c203a86 : SUCCESS in 19m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f3627dd78d694fa9af2159542cdf8bc9 : SUCCESS in 55m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/00a911338f1445028295a128cd820ed9 : SUCCESS in 1h 28m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/862d50e4cfcc4a14b08e11d0fd9466bc : SUCCESS in 46m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1249ed4706fd4397bddaff60987bd110 : SUCCESS in 1h 27m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9d9928ac8c8c4d3b8eef85bb1031b1a8 : FAILURE in 20m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f07931f29fe54b86bf88e20331742e23 : SUCCESS in 1h 43m 54s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3af66e354a54136a000289a8cda65d9 : SUCCESS in 1h 23m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6c8be73f6d64e4caa679659073f7860 : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":4},{"id":"f5c08c22ca9b75699863db24a5fe50115315697d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-16 12:01:41.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"c2dde3fd28b9f5163a9b12383bea1f4a99799cc6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-16 12:10:52.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/80e214c4d9004749a90fccf94780dacc\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e91f79543b274195a4b7808ef8d1cbcf : NODE_FAILURE Node request 200-0000068063 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d847c4954a6e4a33981b909b479f0834 : NODE_FAILURE Node request 200-0000068064 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/666e99b8d5044ce485159232a8dfeb29 : SUCCESS in 46m 16s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/966ae54eb8f84bb283b12e690bc07a0b : SUCCESS in 43m 32s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/90e7885f0f574e8696a8015e3e3c37d3 : SUCCESS in 44m 45s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ae99f68b11a542dcb90a35bf051bcac8 : SUCCESS in 1h 00m 37s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/90fc0c5cf47b43248d22b6de25b08b37 : FAILURE in 1h 54m 40s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a00bcfe6b97e4863ada006545cac9068 : SUCCESS in 46m 40s","accounts_in_message":[],"_revision_number":4},{"id":"661fbe3e6249e2412b829f051e0eeda86fb4c953","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-19 04:38:40.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/4/20250619090241/NEC-V-FC : SUCCESS in 4h 28m 12s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/4/20250619090241/NEC-V-ISCSI : SUCCESS in 4h 35m 56s","accounts_in_message":[],"_revision_number":4},{"id":"1f25df818e47b61735053da45796d1068395a8f0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-20 08:05:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/4/20250620080905/NEC-FC : SUCCESS in 6h 17m 00s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/4/20250620080905/NEC-ISCSI : SUCCESS in 8h 56m 20s","accounts_in_message":[],"_revision_number":4},{"id":"983c1591cfb3a1833aac0c0a29ad010a32f13420","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-20 16:05:03.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"511468125ec49f18860edddea0d56ea75a512680","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-20 16:06:08.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4ddc1c6597904dd29ad18a0444c5edc3 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1c17d64b50404447837f2d5b6788d322 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"e4616b3fd6bcfb2a2f0e9c6cf50a8ddf640f05da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-20 16:09:03.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"95b1f9d456e3186873a306ee9c2aa0e60edde6a9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-20 16:10:09.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8ef2cc8c74044eebab2fe1eb8050de40 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a2a710f53b2d4c679f9ff770364e050e : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"7a1c13fea8a687b29f8a5aa99126d8bb38d43d61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 16:27:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39a7b0b315494d4d8670e660c8a71f6d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a92fd4ce0d4c471d86150ce5daa2321c : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d57bd8cae8df436b9db901af9df6122d : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd9417f7065c40e3b7ce4dd4ae7d9fc4 : SUCCESS in 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"74b265ecb47b9436aa2b5eb59b962a036047fa4c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-20 17:02:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950927 : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":6},{"id":"5fcc0814d4ecc38d08308921c696c2c252835654","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-20 17:04:01.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e29df5db34eb405da75d5f708cf59061\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c6e623323e549f882fb9864b5bc710b : NODE_FAILURE Node request 200-0000068657 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2f9cfdaa534244ae97159612a5bc2806 : NODE_FAILURE Node request 200-0000068658 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd2817696415450c890847f4711e4397 : SUCCESS in 48m 29s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/11133f42867a464db93960438e8c2343 : SUCCESS in 54m 27s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5d7596f267e74869bff38794d00380fa : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":6},{"id":"5cc41626d19b780c1458205cec71b2df10c9be08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 19:15:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b623822b12bc42b39f7b9943ab607d2e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/665961f5e0c4491891a35078d4057b8a : FAILURE in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ebcf833f211421c9399d9c4f1a5fb68 : SUCCESS in 11m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65cadbd7d40a487ea5692e3a3144421a : SUCCESS in 10m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a6b083548dd4894aabece613123f6b4 : SUCCESS in 16m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f0ddd948b474e189bff219c67629c49 : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/881a61189195486782fca6f64fa3fb5f : SUCCESS in 51m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cdbc0933b48b4f63b390f1b9ec6ffe88 : TIMED_OUT in 3h 02m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/767826c883a8420dbc184a56f86a0d2b : SUCCESS in 1h 20m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9480a6a1d5e94a72a956fef61b685017 : SUCCESS in 51m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7e9f96e62f24ad29d42210410090909 : SUCCESS in 20m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/12dccfe8f14c401e8f29d9a2d351ebe8 : SUCCESS in 23m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6c176d472775425d86f47495c6e726f0 : SUCCESS in 6m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ae22775194024f93ba5efabab9c3f01b : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/07b3762729f446ecb4e9c9d74bde96b0 : SUCCESS in 6m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bcfac5dc6da445c4b9d32e62920e1793 : SUCCESS in 8m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/53fdfee73add4c1ab2b9adf142d494e8 : FAILURE in 28m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5b96ff2f5d541e8bd0ae6c03124b3fb : SUCCESS in 7m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/353018343a9c447ba0362472645225fe : SUCCESS in 1h 52m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/520cdb1d426940098082c09095cfa738 : FAILURE in 2h 25m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bbc17de787b8435391fd0d2538803130 : SUCCESS in 1h 55m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4183a2248f584d16a9f910bc8a98f4ae : FAILURE in 21m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f80f87bd222c4dea844a53712c1de372 : SUCCESS in 56m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eb03850dd8354930b68a6d8d665f87bd : SUCCESS in 1h 12m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4635e34ff36a44138da4a7a2d76faae3 : SUCCESS in 1h 14m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38da963438e947a688a00c188f092085 : SUCCESS in 32m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bb4d30d4621c4f9b98110a5af120f3b0 : SUCCESS in 1h 34m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c5fc583a7dc9421a80cfa33781a5484a : FAILURE in 13m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29e5e078c4274321a146c10880c16ba7 : SUCCESS in 1h 27m 16s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34aa298b0eca4f3799a2694bd6b0dca0 : SUCCESS in 1h 37m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76348d1365ba4080abe75fb091c71918 : SUCCESS in 57m 24s\n\nWarning:\n  Comments left for invalid file cinder/api/api_utils.py\n  Comments left for invalid file cinder/api/common.py\n  Comments left for invalid file cinder/cmd/backup.py\n  Comments left for invalid file cinder/cmd/volume.py","accounts_in_message":[],"_revision_number":6},{"id":"faf958c801e7cc97b6a3f4707ce7e9660256cc15","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-20 20:36:38.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"fa1a1569f963651b5390c82eaaee182b977e6d7d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-20 20:39:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1148/.","accounts_in_message":[],"_revision_number":5},{"id":"6f3337140a2b99ca0a209c6cc5bcbbd1ff3cdb84","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-20 20:50:21.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"d5f80b3e2073421d07e103ad1c7b73d21d826772","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-20 21:32:01.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1019/.","accounts_in_message":[],"_revision_number":5},{"id":"436c43212d701fd9f451b4c8690fc4af9351895a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-20 21:34:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1149/.","accounts_in_message":[],"_revision_number":6},{"id":"b55a84f2195a58e93b46158710cbf177eadbb217","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-20 22:46:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1020/.","accounts_in_message":[],"_revision_number":6},{"id":"24f3e969a6987349eeefb969cc6d1eb6d5a6c832","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-25 05:14:56.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/6/20250625095016/NEC-V-ISCSI : SUCCESS in 4h 15m 03s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/6/20250625095016/NEC-V-FC : SUCCESS in 4h 24m 37s","accounts_in_message":[],"_revision_number":6},{"id":"483becd423e1d1b5fde5e587ac7a597fef2b7c0a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-26 08:02:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/6/20250626075340/NEC-FC : SUCCESS in 6h 15m 17s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/6/20250626075340/NEC-ISCSI : SUCCESS in 9h 09m 05s","accounts_in_message":[],"_revision_number":6},{"id":"63bc75d2d4f90a58dce528a5ac741f99396b8963","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-06-27 19:15:46.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_30555\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}],"_revision_number":7},{"id":"280f7bbabe50b5b78753de35c1aa2bbd5adf935a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-27 19:16:40.000000000","message":"Patch Set 7:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":7},{"id":"99bd8e854ffa5d596f7ac3653062478e09df39dc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-27 20:11:37.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2a9863f949934d61bc93da2e05ad7c23\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f7028f0f272e4bfeb0ae83e44d3ca33d : NODE_FAILURE Node request 200-0000069689 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d732edb92b8c4d9db8fcb44b02779ea5 : NODE_FAILURE Node request 200-0000069690 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/97c8ab49dc774d9f83bb8e7e7baa5dbc : SUCCESS in 48m 49s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6103f04a5ef941f9a096222cf048b87e : SUCCESS in 51m 44s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/591454aa7f7a4b57a577d590a13c6852 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":7},{"id":"c88304c3a6fc1e770c6053104169ae3c933da44c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 20:26:59.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b67929a51fef4997811d9c3ce05d6ac6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b9da919d96344460a91f4e1898c4ba2c : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4a034ff5e4b644119bcdbb3c439b6e58 : SUCCESS in 26m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d1c8d8327f7b4a319893a0909bf8ecb1 : SUCCESS in 28m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d48f88e0cd553b1b936fb8d70d29ad89b03752c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 21:38:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d570e0d93a134f08a9eab6bb724d951d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec211912690943c2b7f6d06df28a4826 : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c615074d69e457b983ea8ba3911ff94 : SUCCESS in 12m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9006de30720b4cf7bc2a924e2177aefc : SUCCESS in 10m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f1e862eba65a44b38a7e315e9bac42d3 : SUCCESS in 12m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b5a50645abd48749f14448a82aa58b0 : SUCCESS in 15m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/52dcbba3671549f4b1b2632b1cbe2f67 : SUCCESS in 49m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/19a7ec24f94746cc849cb4d4b3d01e86 : SUCCESS in 1h 03m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ac277cac76b54fcebf3a0a2418961233 : SUCCESS in 1h 31m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7731a3c4b174b899a1155f82d5be25c : SUCCESS in 58m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f539921ee2e4240919c7950ab9d1d76 : SUCCESS in 10m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/384a3cd279144c279795de8fe2d75830 : SUCCESS in 23m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e0e38f5df522445687d9194b156bea5c : SUCCESS in 4m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03bb9ab7ef204f91841dd115383fd3fa : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dc92e05a8bac4966b80cd673186f4ed6 : SUCCESS in 6m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/75c2e6a342014497a10f543ddea3a2ab : SUCCESS in 7m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ecf6f6901cba4a55956470d8b8e3c33b : FAILURE in 18m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/079d3e122bda40cea6d0837022f7055a : SUCCESS in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/02517bb8d8224694af1fbc7066e19ed7 : SUCCESS in 1h 59m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/12b04c7ac2404a81952514bce94bb57d : FAILURE in 2h 13m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6dc662579c542dab659619db394d5f2 : SUCCESS in 1h 53m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4a16e891bf8247c389dd97d057b2d2b9 : FAILURE in 9m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d2a1d80026114defa43b945f915bd9aa : SUCCESS in 32m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aef00653e4234490b00c5aecaab61193 : SUCCESS in 57m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c07967e2f0b0481a8479e274b0305ce7 : SUCCESS in 1h 35m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0e358225df784f11985737b8390099dd : SUCCESS in 26m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1de2f5ab4b4b42f49aa850aa6ce57136 : SUCCESS in 1h 30m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/44e0df9e8e084381bb6799aa9c03791a : FAILURE in 8m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/feaf25d3dbfc408f806506f346c8254a : SUCCESS in 1h 37m 00s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72afb2f32f43449dbaf343f42f5f6c50 : SUCCESS in 1h 54m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21fdad9ed02e42709cf9024f68b681d4 : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":7},{"id":"c47efbeb36b5da45f2960b5cd8d7b46905dc875d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-06-27 21:40:02.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"1074bf94ddeb15ea4b812f0d7688d405e0167aaa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-27 22:44:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1099/.","accounts_in_message":[],"_revision_number":7},{"id":"d0ccffd8c76e2fa8e72aad16c497c0a4631b5e5b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-27 22:52:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1228/.","accounts_in_message":[],"_revision_number":7},{"id":"14ff9179fa877e8c00205aa82527b72c39c37ae0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-28 00:03:46.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"e9abe8bcbcf24e0dba73e6d14d009812db024069","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-04 06:49:03.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/7/20250704093313/NEC-ISCSI : SUCCESS in 5h 47m 03s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/7/20250704093313/NEC-FC : SUCCESS in 6h 15m 48s","accounts_in_message":[],"_revision_number":7},{"id":"1de0006c2f2b62a8f64c86f9e15feef6e118a3de","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-11 01:26:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/7/20250711042347/NEC-V-FC : SUCCESS in 5h 11m 49s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/7/20250711042347/NEC-V-ISCSI : SUCCESS in 6h 02m 30s","accounts_in_message":[],"_revision_number":7},{"id":"89eb10989592f110f2c51c3c50e90cabd12618f9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-16 20:40:36.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"26dc477376d76293065d6714c8f943cc9b7ba433","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-10 11:57:28.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"3b2a66de3487a17770e0ef423ea1cea2e5f428ba","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-10 11:57:40.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"851f571761354e6e16f24bfc20b693fdab0989c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-10 12:04:02.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"9b1dc3723e74fd71065c5453607906fbb5fe5307","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-10 12:29:04.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"64b63fd38d9c106adb51ab12c155cfd38f06baaf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-10 12:34:53.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"7698301f6121efb86880cd2cf7257f651230c416","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-10 13:26:59.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibf260fe254b048d9aeb038c388018b26293f8d8a .","accounts_in_message":[],"_revision_number":8},{"id":"f99b1ac930fe86578fe7f013336bafe80f44f610","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 13:29:59.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e18e803959145709b4d026af4a847c6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e23ed7297db048528c30a4564b132ab1 : NODE_FAILURE Node(set) request 370bb1a381734a0ea6eeaf682eb8d60c failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe86712e83df41c9806d71a5bb657133 : NODE_FAILURE Node(set) request d5c51f38645f4a2096ecccadcea1371a failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce843ac0184947ff93ea098e9c3f777d : NODE_FAILURE Node(set) request be380866b4b04258bbc08585525c59c2 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a55d65aee9efb017bf01d03cd69f64f569e89a92","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-10 14:18:35.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibf260fe254b048d9aeb038c388018b26293f8d8a .","accounts_in_message":[],"_revision_number":9},{"id":"943e268d2e4b684823ec680c2afa279e19a10fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 15:30:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7480302b91f540258d6ab98578328b64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9319997dc939413f9418dbdb04bc2760 : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/362bdee856ef4566acb533e9194e3fcf : SUCCESS in 11m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6cea0dab6bef42cea158ef40a840210b : SUCCESS in 11m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0d68805215b4c2384740041d2b69524 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0cb1bc6b1c94f26a0f6f6a8c29c124a : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/ab5a15e40145488eb36c555aa9dc6863 : SUCCESS in 34m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1019419d2c074b77b1e13e6e2e201ddb : SUCCESS in 51m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d38f26fcea7465cb01d61b2ca1fbab6 : SUCCESS in 1h 41m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8547607b9cd48cf8728a7068b6b1da4 : SUCCESS in 59m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30ed57ed3a1b4acd81ff05531375278f : SUCCESS in 9m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8888ed0a8f8b47a9b4a8830f42f829af : SUCCESS in 27m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4b73c2a4b4cf4327a5795c210c8b4864 : SUCCESS in 8m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/314d622f044b4541a791e5e9b9c37f5f : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fdaad61f3078456a89bbed724853d07e : SUCCESS in 5m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6400d38862ad49c2a9ee37f157ace3a0 : SUCCESS in 8m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9379906d4a0f4a74b804262e0cbb40c5 : FAILURE in 39m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/91666c6ab7a84631835710510caf0420 : SUCCESS in 4m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/61c75ee2999b4429900a2db4f72ab091 : SUCCESS in 2h 09m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/318a3bd780c54c099f5dfa517c27a21d : FAILURE in 2h 39m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/10827583722c4b50b63bf8fbc1c52754 : SUCCESS in 1h 57m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/47abddac93604ef0ab9f8a85386f82d8 : SUCCESS in 1h 54m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f74f268e578d4c34b59e7ebbaa2e89f8 : SUCCESS in 30m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a2b00cf40f68494ba5b57c20fd9086e3 : SUCCESS in 1h 13m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2d781f38be254c86b73c224d575f62be : SUCCESS in 1h 35m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ad4e3da7702049eeae50756427d99730 : SUCCESS in 39m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7a370da8333d4892abafa042d5b4ae2b : SUCCESS in 1h 09m 17s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/59b102701c8249d18203acb07579f57c : SUCCESS in 1h 50m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1586742498534232b4234969aa37f951 : SUCCESS in 55m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53041a0524094fcb81d476c0c9af5635 : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":9},{"id":"c10d89fa7b5dc3b45c6fe0b21babfdbc2a5df833","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-10 15:36:46.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_10_2025_15_30_27\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"f2e781b52aefa79d548fb8df082cbca171bcb032","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-10 23:04:37.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950927 : FAILURE in 1h 11m 25s","accounts_in_message":[],"_revision_number":9},{"id":"7d5507f0b3a785b5313a46ff194f7322e0ec02ce","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-11 03:01:32.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b3e04bcf1e794ef99f286f2090ca0bda : SUCCESS in 1h 36m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/470de2985f2c46f9b7fd0daeb503ba8a : FAILURE in 1h 47m 17s","accounts_in_message":[],"_revision_number":9},{"id":"92989a36f5d10661c260b691ddd83ebe0654ee16","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-11 20:22:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2001/.","accounts_in_message":[],"_revision_number":8},{"id":"87b11b40a514a6ba74752104093b7564c4c316ed","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-11 20:45:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2115/.","accounts_in_message":[],"_revision_number":8},{"id":"6fda7bc8f66bed5f6b3324a72bac95fba8a1e13a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-11 21:20:26.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2002/.","accounts_in_message":[],"_revision_number":9},{"id":"c876e82fac90fe3d05486c429b4067689dccc29e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-11 21:45:55.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2116/.","accounts_in_message":[],"_revision_number":9},{"id":"331b3c655d6e195b5fd2d8c65183ee961b90b2e9","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-20 15:20:12.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"472f0fd5e7b70aa05c855eaef63de2898d17d060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-01-30 20:11:48.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"02c1c6baf43944a45c7243dd400638f51daba35b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-30 20:11:55.000000000","message":"Patch Set 10:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/767/","accounts_in_message":[],"_revision_number":10},{"id":"26441002398872c7d9570f975a79b182b145f643","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-30 20:11:58.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950927/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950927/10/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"c3052460dafeca449212592630871e388ce851a4","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-01-30 20:12:39.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"db0f4738a2afca34a9a022688d1f75a3c5397446","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-30 20:12:50.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2dafd5d5b7d04ba7ab6c0ba5a10efe7a : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c53209196e3548929b7dc6b92d4401b0 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"8dac54d7610093c5f10980ef54dc82af5d7f957d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-01-30 20:28:57.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"600a90f5120939e851ff4754d751ef87c15bdea9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-30 20:29:05.000000000","message":"Patch Set 11:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/768/","accounts_in_message":[],"_revision_number":11},{"id":"33a20fd18513379adb4ef2691f4ff54c08ee7f32","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-30 20:29:09.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950927/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950927/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"ada3de6866665fedbf5f4da008ff7468cffc61a1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-30 20:30:00.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8050af5e2b704dfb90516eea120981a7 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4e0d6face5d542a3b40b3facc874ae99 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"a8f958cd4310f4702e2119e00b1b3c83c061758c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-01-30 20:30:58.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"c18d90beb5292b0a2b71c97219c2b263406a4a7a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-30 20:31:10.000000000","message":"Patch Set 12:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/769/","accounts_in_message":[],"_revision_number":12},{"id":"c02dd082cab45f9fa498b66476d6c098accbf29d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-30 20:31:14.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950927/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950927/12/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"dc645a6d5c974c434af3f36cbf8fa304b7d4b22c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-30 20:32:00.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/17212e3cde9b421f936b921f3eb8b62b : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/180e538a41244857a6bf1a3115499b70 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"d0d9c3cc20104ddfc70a24ae98c7bb1edcd3b8e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 20:50:43.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5f1c23a247d4b7da90c500c595ea882\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/63dfda6afb264de3b275f2952ebedaad : SUCCESS in 19m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8df2fff616d042acbcfe78a5d8d673d4 : SUCCESS in 19m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1d77bd5458f14384a8ed7759e352d0c2 : RETRY_LIMIT in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"06f0446f929db1c43117402cf89ea02b3d405fbe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-30 21:12:58.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibf260fe254b048d9aeb038c388018b26293f8d8a .","accounts_in_message":[],"_revision_number":10},{"id":"487ca6b3431ec432a9f882cae2f05eec8a4cbb98","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-30 21:29:00.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibf260fe254b048d9aeb038c388018b26293f8d8a .","accounts_in_message":[],"_revision_number":11},{"id":"d08378e02c1dae7f70f6383e63185ab5ceef2093","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-30 21:31:48.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibf260fe254b048d9aeb038c388018b26293f8d8a .","accounts_in_message":[],"_revision_number":12},{"id":"910db7e0a841f0e73c349398b1eef9af9d201313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 22:42:16.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/761703a5bb5f455590d232e024b409ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3b6bf779634ea68c76d9838a262b98 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b27f6694a054c34944a33d36adf6c3a : SUCCESS in 11m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39670b997d61471e805e603e3d2200f3 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610a6edb562a4999afff28233c6797ea : SUCCESS in 13m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/d94c43ae409444ee8cb92ba46a5c117c : SUCCESS in 1h 00m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/074a9dfb50b74d4f8bf86aaf9be150ee : SUCCESS in 28m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5827c951351d43f59c94fa7d38158847 : SUCCESS in 1h 30m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aba929b9951840be85b36034529912dd : SUCCESS in 1h 06m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ecaf05734314d0383d80fc564456527 : SUCCESS in 17m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dd475b59ff5649279ceda43a48620457 : SUCCESS in 20m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/410e7a421f2746c8b77f1673735696c6 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/96fc3eef6e7f42679aaf5e1ef083c62f : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7e8b7c6a994b4e268426bc83b792c54f : SUCCESS in 5m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b28bb34f22724cdbbed03d9c78f92672 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fdb0d348be3d46d685f9c311f60463b3 : FAILURE in 38m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/29fca1fbd0034ca2830be8433f096eff : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/26d6eff1976a4ba6abacac00310f405a : SUCCESS in 1h 27m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fe320a8b8d3846b088dff2d5fe75e349 : FAILURE in 1h 30m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9c58bbecdce5418d96e32b6441ec5b3f : SUCCESS in 2h 02m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/dabafe76d0594a7594a26b9779103b16 : SUCCESS in 1h 29m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/30d87ed833bc488a84236111539e5cf9 : SUCCESS in 35m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0984518cfd39487e81ab3e3ffe4965e8 : SUCCESS in 1h 08m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6e68c6cbb874bd58f856f28bf427fb5 : SUCCESS in 2h 04m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9bc81cf5a6244d909950eb9bc30de851 : SUCCESS in 47m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/54b037877e374b2ba63062053bddd684 : SUCCESS in 1h 00m 31s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/226acf85593c424db564a5437c1d7de1 : SUCCESS in 1h 58m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e69177db9fb8447fb26689256381758c : SUCCESS in 52m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/611bace974e645e08aa2ebe397bb5b98 : SUCCESS in 30m 23s","accounts_in_message":[],"_revision_number":12},{"id":"382b39d6b1e0cb8f3a359b0fa4a11223e0d40f58","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-30 22:44:08.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/eb6ad67f75a94c99a14d4d524b59e5ae : FAILURE in 38m 33s","accounts_in_message":[],"_revision_number":12},{"id":"8df1cb508c6a9464ea8870c88f0dbf8ee19bce5f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-30 22:45:05.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950927 : FAILURE in 39m 39s","accounts_in_message":[],"_revision_number":12},{"id":"42b6d71f81b53669991489268b4a077a431fc94e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-30 22:49:14.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_01_2026_22_42_26\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"f23502c870137c14ab5866ea7b8233c4fac0a744","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-03 00:24:50.000000000","message":"Patch Set 10:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_461/index.html : SUCCESS in 92m","accounts_in_message":[],"_revision_number":10},{"id":"a863831d83a17e15ca43c2eba61256a6fb3ed713","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-03 02:13:10.000000000","message":"Patch Set 11:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_462/index.html : SUCCESS in 108m","accounts_in_message":[],"_revision_number":11},{"id":"ea5da41ef5f0005d2ad8dc9a187627c8286c4f4d","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-03 03:50:40.000000000","message":"Patch Set 12:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_463/index.html : SUCCESS in 97m","accounts_in_message":[],"_revision_number":12},{"id":"d97b4406a6c9ad6a4907d766e9c42d8d870a6a3f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-02-08 07:18:51.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/12/20260208060009/NEC-ISCSI : FAILURE in 9h 52m 49s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/27/950927/12/20260208060009/NEC-FC : FAILURE in 10h 18m 40s","accounts_in_message":[],"_revision_number":12},{"id":"f6b8fedabd9337274f8041470fdfc029b4489ab3","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-10 19:39:21.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"a216845f9246383d53258297591a7c1099060cca","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-03-11 13:32:07.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7a41798744c6286817f9aabf5ee3bac92a18e0a9","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-08 15:33:01.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"2d89504eed18e651dc491c01d57829abd4a2113d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-05-28 23:24:31.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"475c7c84ca380ffa3f8c5ee89a35c8bc3f5562c3","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-05-28 23:33:25.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"acc7cea565aa082e449e2113ed0e252e629bf1d5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-29 00:05:19.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/fc44311f08c84b0191837c8e821c29e3 : SUCCESS in 40m 10s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/950927\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/950927","accounts_in_message":[],"_revision_number":13},{"id":"88e432be18af1fa4b9a499bac543129a89078a2f","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-29 00:07:17.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- hpe-fc-driver build/e0472526288d4eb8a0b168be560f6fb0 : SUCCESS in 42m 14s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/950927","accounts_in_message":[],"_revision_number":13},{"id":"a8cb4622b065b44b8c8be46065accf59fa8d5085","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-29 00:10:59.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/601604ccc53041a2a389b6d17ed651da : SUCCESS in 46m 02s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/950927","accounts_in_message":[],"_revision_number":13},{"id":"38dbc58d21821da37ec5618d9ef5f5766e56a047","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-29 00:21:58.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibf260fe254b048d9aeb038c388018b26293f8d8a .","accounts_in_message":[],"_revision_number":13},{"id":"171e72d57de1f98efc1a7d81a620ac96c6a65dae","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-29 00:30:52.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest build/9d9e3430a90d4e63a617c77adb6ecf4c : FAILURE in 52m 00s\n- cinder-storpool-tempest-iscsi-multipath build/d07e6b82f73746619e40a18abe960fb5 : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":13},{"id":"7ccd7aa7a23721498d7ed5d2ed22221c47a1aace","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 00:33:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4276/.","accounts_in_message":[],"_revision_number":13},{"id":"420d3b27d0b21e72b2a6987cc63787870d0a7c94","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 00:38:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4100/.","accounts_in_message":[],"_revision_number":13},{"id":"92d9a83678bdc160da973ca6c8d28205eae90e19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 00:39:06.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6459354722c24d8383711e3e0b866ad8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f904ba43f226438daf152eea92b91655 : SUCCESS in 16m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d0ff8d6670ee40d0bf560868609514d2 : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b439279175b148dc952c64c284b8f6ff : SUCCESS in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"796ab3d4d34aa849252fb96decb6932606f14c96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 02:11:57.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c86e195019d49cf940e63f638b42d97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ad5c3d13f754bd6b5ec61871999e975 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7177e386e9cb48fc9d0dc6f66476f551 : SUCCESS in 11m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfbbe59eca034f4cbf0538d4d4319ed3 : SUCCESS in 8m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4b4f43193bc8463e8a016a70c3056c5b : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6340c7450c04ed1a1c64e3e0ad863aa : SUCCESS in 15m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/3420e65e27d34db1bb0a98440e2157a9 : SUCCESS in 35m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f74b63f0d21415898cda6c1daa0a238 : SUCCESS in 1h 07m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/441385867c1d4c9389b8014a5a566459 : SUCCESS in 1h 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e18e7b8bb3f428c9d098cedf97c9cf0 : SUCCESS in 1h 04m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4ef6c7ad0e54a98ad2b303d0b57b27c : SUCCESS in 8m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8af2db0c44c2435abd364850782cb8ba : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd3bc897e49e433aad456e4b05853482 : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4f9b1a4f31e4489692b1520ec56b25d1 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7cf0ef3898c040a392ee01e08641429c : SUCCESS in 4m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c074b6a402964c7685b9832a77c9740f : SUCCESS in 5m 04s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/982bb6511fc44ccc9aa7ba514a619247 : SUCCESS in 8m 21s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/40f82cbbc0e1445ba84f133e9866e29e : SUCCESS in 39m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93299810e5584d2e92260789ab9cc9e7 : SUCCESS in 3m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f5fee1f3617475bb8908671bf7e642a : SUCCESS in 1h 58m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3d7dd2fca0be46f0b6d3d098245d17aa : FAILURE in 2h 42m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6726980c887436faa2c7a021668c98a : SUCCESS in 1h 53m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6cad164a847a43398c33dc3b5b808e8c : SUCCESS in 2h 03m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5ba19188d3e9461092d8b4cbc2b378a5 : SUCCESS in 39m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bcbb882c0ecf40db807ea3ced6e87756 : SUCCESS in 41m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8f80ae6e96e84774b37f58e8321cdfeb : SUCCESS in 1h 33m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/37341db97f404e4185b894b7966c2260 : SUCCESS in 42m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf9e150b19ba4eeea8c06e5854e75822 : SUCCESS in 57m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92d66eaca52d4c3ba4dc05028516261b : SUCCESS in 31m 11s","accounts_in_message":[],"_revision_number":13},{"id":"fed82b7e6865d5e6c3b090845faac659b871eda1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-29 02:22:19.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_05_2026_02_15_12\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"820753ffa2e4ce26c12cd148c524cb8c9958c3fb","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-29 04:18:10.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-27-950927-13-fc : SUCCESS in 1h 8m 53s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-27-950927-13-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"f643e10e6de1791f81307cc08ebde7e603ee1735","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:49:41.000000000","message":"Patch Set 13:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/2001/ : ABORTED","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"700efe6aa3ba321b4cf11b3db6e434e0b87c31c0","revisions":{"51a04546f59528696fbc163f39d50ee3d255c6c9":{"kind":"REWORK","_number":1,"created":"2025-05-27 02:12:25.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/27/950927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/1"}}},"commit":{"parents":[{"commit":"3370f905796b0e2e1cfd7a121290e5313bc8f6b6","subject":"cinder-manage: Use same timestamp for purging deleted rows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3370f905796b0e2e1cfd7a121290e5313bc8f6b6"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 02:11:53.000000000","tz":-180},"subject":"[WIP] cinder-manage: Perform database purge operations in batches","message":"[WIP] cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the `innodb_buffer_pool_size`\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional `--batch-size` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51a04546f59528696fbc163f39d50ee3d255c6c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51a04546f59528696fbc163f39d50ee3d255c6c9"}]},"branch":"refs/heads/master"},"2faa6ecf7a4b0328d56d2f65d4bfd4a5c7c763e5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-27 02:13:56.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/27/950927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/2"}}},"commit":{"parents":[{"commit":"3370f905796b0e2e1cfd7a121290e5313bc8f6b6","subject":"cinder-manage: Use same timestamp for purging deleted rows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3370f905796b0e2e1cfd7a121290e5313bc8f6b6"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 02:12:54.000000000","tz":-180},"subject":"[WIP] cinder-manage: Perform database purge operations in batches","message":"[WIP] cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2faa6ecf7a4b0328d56d2f65d4bfd4a5c7c763e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2faa6ecf7a4b0328d56d2f65d4bfd4a5c7c763e5"}]},"branch":"refs/heads/master"},"f3b4530e73a1203aa2dc228945b51d0c05ad7a97":{"kind":"REWORK","_number":3,"created":"2025-05-27 15:53:55.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/27/950927/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/3"}}},"commit":{"parents":[{"commit":"3370f905796b0e2e1cfd7a121290e5313bc8f6b6","subject":"cinder-manage: Use same timestamp for purging deleted rows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3370f905796b0e2e1cfd7a121290e5313bc8f6b6"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 15:53:55.000000000","tz":0},"subject":"[WIP] cinder-manage: Perform database purge operations in batches","message":"[WIP] cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3b4530e73a1203aa2dc228945b51d0c05ad7a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3b4530e73a1203aa2dc228945b51d0c05ad7a97"}]},"branch":"refs/heads/master"},"7355b4083018c823e6ee1e333c5702932ce03fa0":{"kind":"REWORK","_number":4,"created":"2025-06-16 10:15:40.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/27/950927/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/4"}}},"commit":{"parents":[{"commit":"111e43b9c05c3a7ac8e14771d976bc6c68a70d9c","subject":"Merge \"tgt target: Provide unique scsi_sn and scsi_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/111e43b9c05c3a7ac8e14771d976bc6c68a70d9c"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-16 10:07:25.000000000","tz":-180},"subject":"cinder-manage: Perform database purge operations in batches","message":"cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7355b4083018c823e6ee1e333c5702932ce03fa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7355b4083018c823e6ee1e333c5702932ce03fa0"}]},"branch":"refs/heads/master"},"c8185254fa45bd4388c3990f49a2c524b1f21390":{"kind":"REWORK","_number":5,"created":"2025-06-20 16:05:03.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/27/950927/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/5"}}},"commit":{"parents":[{"commit":"111e43b9c05c3a7ac8e14771d976bc6c68a70d9c","subject":"Merge \"tgt target: Provide unique scsi_sn and scsi_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/111e43b9c05c3a7ac8e14771d976bc6c68a70d9c"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-20 16:04:43.000000000","tz":-180},"subject":"cinder-manage: Perform database purge operations in batches","message":"cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8185254fa45bd4388c3990f49a2c524b1f21390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c8185254fa45bd4388c3990f49a2c524b1f21390"}]},"branch":"refs/heads/master"},"6e830b02dcd3f2cb73aaaa6255af0c43f4a037b5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-06-20 16:09:03.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/27/950927/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/6"}}},"commit":{"parents":[{"commit":"c948b22eace9f6aa299a0af3ac374864ff8ff163","subject":"Pin flake8_import_order version to \u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c948b22eace9f6aa299a0af3ac374864ff8ff163"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-20 16:08:37.000000000","tz":-180},"subject":"cinder-manage: Perform database purge operations in batches","message":"cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e830b02dcd3f2cb73aaaa6255af0c43f4a037b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e830b02dcd3f2cb73aaaa6255af0c43f4a037b5"}]},"branch":"refs/heads/master"},"1d5ae9814f834d6c0d5e52192c58907b140835fb":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-06-27 19:15:46.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/27/950927/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/7"}}},"commit":{"parents":[{"commit":"53245bce31549a29ce74dd49de6c9f7417868126","subject":"NFS driver: Fix fail to resize NFS volume with snapshots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53245bce31549a29ce74dd49de6c9f7417868126"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-27 19:15:46.000000000","tz":0},"subject":"cinder-manage: Perform database purge operations in batches","message":"cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d5ae9814f834d6c0d5e52192c58907b140835fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d5ae9814f834d6c0d5e52192c58907b140835fb"}]},"branch":"refs/heads/master","description":"Rebase"},"5d3b8c36b4095e768c48d407221e1a9293325e65":{"kind":"REWORK","_number":8,"created":"2025-10-10 11:57:28.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/27/950927/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/8"}}},"commit":{"parents":[{"commit":"53245bce31549a29ce74dd49de6c9f7417868126","subject":"NFS driver: Fix fail to resize NFS volume with snapshots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53245bce31549a29ce74dd49de6c9f7417868126"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-10-10 11:57:21.000000000","tz":-240},"subject":"cinder-manage: Perform database purge operations in batches","message":"cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d3b8c36b4095e768c48d407221e1a9293325e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d3b8c36b4095e768c48d407221e1a9293325e65"}]},"branch":"refs/heads/master"},"292d71a43ba139e29019e6c3b75feddb4bc9dfa2":{"kind":"REWORK","_number":9,"created":"2025-10-10 12:29:04.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/27/950927/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/9"}}},"commit":{"parents":[{"commit":"53245bce31549a29ce74dd49de6c9f7417868126","subject":"NFS driver: Fix fail to resize NFS volume with snapshots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53245bce31549a29ce74dd49de6c9f7417868126"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-10-10 12:29:04.000000000","tz":0},"subject":"cinder-manage: Perform database purge operations in batches","message":"cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/292d71a43ba139e29019e6c3b75feddb4bc9dfa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/292d71a43ba139e29019e6c3b75feddb4bc9dfa2"}]},"branch":"refs/heads/master"},"69f31ba8b618ead47a3f79244b25f760d602aad3":{"kind":"REWORK","_number":10,"created":"2026-01-30 20:11:48.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/27/950927/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/10"}}},"commit":{"parents":[{"commit":"47107b4c84b060be834d6f430d32c3ebcac7d35a","subject":"Merge \"Add 512e/4k disk geometry configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47107b4c84b060be834d6f430d32c3ebcac7d35a"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-01-30 20:11:33.000000000","tz":-180},"subject":"cinder-manage: Perform database purge operations in batches","message":"cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/69f31ba8b618ead47a3f79244b25f760d602aad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/69f31ba8b618ead47a3f79244b25f760d602aad3"}]},"branch":"refs/heads/master"},"05812344f8c288c14425a1a4372d72c7d5d80bee":{"kind":"REWORK","_number":11,"created":"2026-01-30 20:28:57.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/27/950927/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/11"}}},"commit":{"parents":[{"commit":"47107b4c84b060be834d6f430d32c3ebcac7d35a","subject":"Merge \"Add 512e/4k disk geometry configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47107b4c84b060be834d6f430d32c3ebcac7d35a"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-01-30 20:28:57.000000000","tz":0},"subject":"cinder-manage: Perform database purge operations in batches","message":"cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05812344f8c288c14425a1a4372d72c7d5d80bee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05812344f8c288c14425a1a4372d72c7d5d80bee"}]},"branch":"refs/heads/master"},"0a33feeb6af8e9e6f467dbed828a73d57289acad":{"kind":"REWORK","_number":12,"created":"2026-01-30 20:30:58.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/27/950927/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/12"}}},"commit":{"parents":[{"commit":"47107b4c84b060be834d6f430d32c3ebcac7d35a","subject":"Merge \"Add 512e/4k disk geometry configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47107b4c84b060be834d6f430d32c3ebcac7d35a"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-01-30 20:30:58.000000000","tz":0},"subject":"cinder-manage: Perform database purge operations in batches","message":"cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a33feeb6af8e9e6f467dbed828a73d57289acad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a33feeb6af8e9e6f467dbed828a73d57289acad"}]},"branch":"refs/heads/master"},"700efe6aa3ba321b4cf11b3db6e434e0b87c31c0":{"kind":"REWORK","_number":13,"created":"2026-05-28 23:24:31.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/27/950927/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/950927/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/950927/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/950927/13"}}},"commit":{"parents":[{"commit":"2ef4fadbca5ed10390926e56bba309cf34b4d4a0","subject":"Merge \"Fix cascade delete for non-admin with deleted cross-project snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ef4fadbca5ed10390926e56bba309cf34b4d4a0"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-05-27 01:34:20.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-05-28 23:24:13.000000000","tz":-180},"subject":"cinder-manage: Perform database purge operations in batches","message":"cinder-manage: Perform database purge operations in batches\n\nThe ``cinder-manage db purge`` command may fail with the error:\n\"The total number of locks exceeds the lock table size\" when database\ntables have too many rows to purge in a single operation. This happens\nwhen the number of affected rows exceeds the ``innodb_buffer_pool_size``\nlimit in MySQL.\n\nThis change modifies how deleted rows are purged by performing delete\ntransactions in batches, reducing the risk of lock exhaustion.\n\nA new optional ``--batch-size`` parameter has been added to allow control\nover the number of rows deleted per batch.\n\nThe Default ``batch_size`` is set to 1000 rows.\n\nCloses-Bug: #2111764\nChange-Id: Ibf260fe254b048d9aeb038c388018b26293f8d8a\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/700efe6aa3ba321b4cf11b3db6e434e0b87c31c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/700efe6aa3ba321b4cf11b3db6e434e0b87c31c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
