)]}'
{"id":"openstack%2Fcinder~950929","triplet_id":"openstack%2Fcinder~master~Ibb4d71868106226b5513e4c825f769008a446727","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-08-08 16:03:06.000000000","reason":"Change was submitted"},"36178":{"account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"last_update":"2025-08-08 16:03:06.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-08-08 16:03:06.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2025-08-08 16:03:06.000000000","reason":"Change was submitted"},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-08-08 16:03:06.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2025-08-07 12:48:58.000000000","reason":"\u003cGERRIT_ACCOUNT_21129\u003e replied on the change","reason_account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2025-08-08 16:03:06.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2025-08-08 16:03:06.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-08-07 19:18:07.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-08-07 14:39:49.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Ibb4d71868106226b5513e4c825f769008a446727","subject":"NetApp - Fixed detach issue for multi-attached volume","status":"MERGED","created":"2025-05-27 05:19:27.000000000","updated":"2025-08-08 22:04:12.000000000","submitted":"2025-08-08 16:03:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":81,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950929","meta_rev_id":"3d8fc3a6b1927f1e89cb8dae4176240dc17a91ce","_number":950929,"virtual_id_number":950929,"owner":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-08 16:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"recommended":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":1,"date":"2025-08-07 12:48:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2025-08-07 19:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-07 14:39:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2025-08-07 19:18:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"reviewers":{"CC":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-27 05:24:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-05-27 05:24:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-05-27 05:24:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-05-27 05:24:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"state":"REVIEWER"},{"updated":"2025-05-27 05:24:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2025-05-27 05:24:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2025-05-27 05:36:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-27 05:53:28.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-05-27 06:06:44.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-27 06:13:41.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-27 06:14:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-05-27 06:40:14.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-05-27 06:44:55.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-05-27 06:45:51.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-27 08:00:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-27 11:01:53.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-05-28 08:52:42.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-05-28 14:55:16.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-31 10:56:39.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-06-04 18:13:43.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2025-06-05 06:17:51.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-06-05 11:08:38.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-06-05 16:08:50.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-06-05 21:09:08.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-06-18 14:05:44.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-07-24 15:07:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-07-25 05:49:04.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"CC"},{"updated":"2025-07-30 06:58:56.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-05 18:58:33.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-07 19:18:24.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"4c76c25bbefc0877b74a92bf13a856cd9e78f62b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-05-27 05:19:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3788fb38b6a4fe9a6556a33ba2604e9d7d7f3d8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 05:36:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd2e105ab54a4bcba56d6cf2e332d197\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c874d2b7f1ff4c85a4f77341bc1a249d : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fda9b761bd5e4e02bc43ff8bc76b9a71 : SUCCESS in 14m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/42bccf570cff4d1f8902f1657b9fe1e9 : SUCCESS in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ae1a0f401d3daf23f62ad67ed6fd97d2aeabf41","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-27 05:53:28.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7cc4e3cae301417bbe08e4dfdd95defa\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7fdd407547184cce8d6999c3fe122fd7 : NODE_FAILURE Node request 200-0000054625 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b29cd30443be48889cf597d38fb318fe : NODE_FAILURE Node request 200-0000054626 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ff1b570a3d27419082876ca6b96d8c19 : FAILURE in 22m 40s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0f4eb9c7963f4378b3e4455c4bd62b5e : FAILURE in 24m 56s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dd3dd338b086466bb78c269f7e9a35a3 : FAILURE in 24m 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f59f50d758f74ebdb7941c1d54a22fac : FAILURE in 25m 43s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e2fc22b717234e60b2172873a9115f62 : FAILURE in 27m 08s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/94bbdbb5a7814339a161424ceeaf0143 : FAILURE in 33m 29s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cd558c0cdebd4a5b8b0348768088950b : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/24ef9aa9dcd34dc782436a6d2d0865da : FAILURE in 28m 02s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-fa-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fa-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#23\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fa-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fb-nfs-noble\n      Collecting job variants for pure-devstack-cinder-tempest-fb-nfs-noble\n      Variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: {MatchAny:{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#35\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#442\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fb-nfs-noble branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:stable/2025.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#93\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#190\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"6512c4ea7ea07acf57a3bfa7b7cd969f388f4e79","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-05-27 06:06:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950929 : SUCCESS in 45m 37s","accounts_in_message":[],"_revision_number":1},{"id":"304765406edcb754d0bcf3a3853a8e1cbcf3fbe6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-05-27 06:13:41.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"56bba99a849c121aacc3248ab6f0466703d3e6fb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-27 06:14:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/790/.","accounts_in_message":[],"_revision_number":1},{"id":"c06728defb27c19dab372932acdd0d085e04f119","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-27 06:17:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/916/.","accounts_in_message":[],"_revision_number":1},{"id":"2f228451312131d55349763e66ef737870232c26","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-05-27 06:40:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950929 : FAILURE in 1h 19m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fcf5c5cc6920d280990e2819dd64d740b10a7bfe","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-05-27 06:44:55.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-1/FC/ : UNSTABLE in 24s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-1/iSCSI/ : FAILURE in 1h 25m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"d29ded9867694e437126aa25d828f2306d05db0f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-05-27 06:45:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e983e17ccf68401c869630c03a3d65a5 : SUCCESS in 1h 24m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f6c49c221b704d1cab101ba172280b35 : SUCCESS in 1h 21m 41s","accounts_in_message":[],"_revision_number":1},{"id":"8b4c7ba16b5cc97c238986d5f7fc1c4fb0a53552","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 08:00:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d2f993e8c3940e0a4fd17da04de5507\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3746ec14c144328a448d2e48113b6db : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bfbe93c3f6d6442c9e0c28d8a63d5d12 : SUCCESS in 10m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e29f3d22fe44f778b2e904c345ef1dc : SUCCESS in 11m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/250158d9c26542ddb5ef2ad7f518bc0e : SUCCESS in 14m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0308b523600f416a82e9a10e2187a51f : SUCCESS in 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/43a0971f6f6446a9a490bc242bc57a9b : SUCCESS in 56m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/547f5dc121fa4a54988edfd6a1f14b54 : SUCCESS in 54m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e2a1f1ab7fe49b5bdb44e62b44cb747 : SUCCESS in 1h 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/527335ddaf084a50905f8e1a24363b56 : SUCCESS in 1h 04m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fee47d86660742f18145f376a95aee51 : SUCCESS in 18m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/137732357d6449f3bddd5ec79258052f : SUCCESS in 22m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/43016905a2d047f483443b385723fc2f : SUCCESS in 4m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c2423a59103406da0b92db36760d270 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0290dbd2d1bd4c64aaa1963a8ea409d8 : SUCCESS in 7m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8c27bb34047740cfb2f0169b2634cec5 : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9bf728e3314a4f17859813902c033933 : FAILURE in 27m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/290ec56bdf124817b37f9a08e40d8c2b : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d98084e365664214b5fd6684a2c09253 : SUCCESS in 1h 39m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/738378a84d534a41bd617b3802dab6da : FAILURE in 2h 16m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/79ef7e7b1f1a438f81a36a0d96c5831a : SUCCESS in 1h 33m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/028fea8fa0014fa681c958c528a0a528 : FAILURE in 19m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5a4646fbc9eb40a98eb25d932fb0619f : SUCCESS in 37m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5501a9537c4e4bd08bd2e4e5c8e833fa : SUCCESS in 1h 11m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3b04ea4afb03437cbd5755223525293a : SUCCESS in 1h 31m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f19b50cc3a5c433c905b1add390fb203 : SUCCESS in 39m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b58f7c2d9acc4136b2c259761c71f7b1 : SUCCESS in 1h 25m 23s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8edc14875fa34eacb5fd30cae3f15943 : FAILURE in 25m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/205bc822a7e84e8e8439945f052e8d5f : SUCCESS in 1h 28m 07s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aed59cf48e274050bd5a6f7eafaf8889 : SUCCESS in 1h 21m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96c79a7a101a426485529086cfbed64f : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6a2db41d59b9606dcb5aa10d85cba4430d7b04e1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-05-27 11:01:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs-rest cinder-tempest-plugin-ontap-nfs-rest : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-iscsi-rest cinder-tempest-plugin-ontap-iscsi-rest : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d79f8140962aa5dd2e13c8a903a709e675ec3ba9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-05-28 08:52:42.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/1/20250528093255/NEC-V-FC : SUCCESS in 4h 12m 07s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/1/20250528093255/NEC-V-ISCSI : FAILURE in 8h 19m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f586d4b9d507858949f851a3629d57750a89580c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-05-28 14:55:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/1/20250528174627/NEC-ISCSI : SUCCESS in 5h 39m 19s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/1/20250528174627/NEC-FC : SUCCESS in 6h 08m 46s","accounts_in_message":[],"_revision_number":1},{"id":"967b4fb32f5ff05ec608546bc10b486bffa7157d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-06-02 15:24:02.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ad76ce5b4e6bc7d75916532c62188e37c59f1599","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-06-04 18:13:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b21f231ab427c3ac72f1123b8d2cff6319591606","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-06-05 06:17:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/950929-1/Seagate-iSCSI/ : SUCCESS in 2:20:46\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/950929-1/Seagate-FC/ : SUCCESS in 2:08:00","accounts_in_message":[],"_revision_number":1},{"id":"a39d32bf5a663da49c672c83978cd39e304f1365","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-06-05 11:08:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/950929-1/PVME-iSCSI/ : SUCCESS in 2:21:12\n- PVME-FC https://os-logs.seagate.com/dell-ci/950929-1/PVME-FC/ : SUCCESS in 2:07:47","accounts_in_message":[],"_revision_number":1},{"id":"8607381f2247831ae35082a2a3fe165bcae585c2","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-06-05 16:08:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/950929-1/HPMSA-iSCSI/ : SUCCESS in 2:25:20\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/950929-1/HPMSA-FC/ : SUCCESS in 2:12:59","accounts_in_message":[],"_revision_number":1},{"id":"3d4a49f162a5134ce63f8318cc3a8234d1521510","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-06-05 21:09:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/950929-1/Lenovo-iSCSI/ : SUCCESS in 2:25:35\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/950929-1/Lenovo-FC/ : SUCCESS in 2:12:48","accounts_in_message":[],"_revision_number":1},{"id":"e9ecaea899715a2848445e3091fb63673f9741a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-06-09 15:48:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b44b4cba9239bf01da205ce690ce636fb00f7637","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-06-09 15:49:45.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4682710d02cca87740d3a1147e8bb037e4107ce2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 17:01:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5566f2c8c43d459ba567e736e487bc80\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/56753b3c59fe468eb9500ad747a2d513 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fcfc8febc9d94a8d944dd73bd1a64cca : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/54206f42d7414a8394b1c475b8e3d592 : SUCCESS in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b360cfe3652861a80ef4dcce7c75e74ba612875","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-06-09 17:13:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f7b27546c9b7b55aac523b5ad35a0374fedd2b56","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-09 17:40:58.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-2/FC/ : UNSTABLE in 2m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-2/iSCSI/ : FAILURE in 1h 52m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"b3ff96382368cd69607799ae93dd508b6349a2e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 17:46:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31bf34732c0e48dbb2b3b0445c7f247e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ebb618a421142678e8a00bd1425a9e8 : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f06ac01160bc4af7a2a77aa3c361c32c : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3375498d52864b798599143f0e0814f4 : SUCCESS in 30m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba9bd426eb4bc6d0f82b854abf4081f3fa2f6039","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-09 18:13:31.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9d18e433a9994fe4968462be8fcf73a2\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5fd09685d0f94df4b9cb6a62cf2defe7 : NODE_FAILURE Node request 200-0000065695 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6e692ec0f18a40e397bc8858bc4fc436 : NODE_FAILURE Node request 200-0000065696 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ba44d3eaaa864667ac6664f97dd4bc32 : SUCCESS in 44m 38s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0781f50d37544e9d9bf57c47c731160e : SUCCESS in 46m 02s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8baa1c1294dc42678d06d66be318e180 : SUCCESS in 43m 14s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a88edc736fc84c50ad535339ece88ac8 : FAILURE in 45m 34s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0d17826d6d264bde83f505ea5ae4ffec : FAILURE in 49m 12s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/067aa128f84a49ba816cbadab51d55b2 : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":3},{"id":"13bc4db91d442250bf9c206b546f26b36a664700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 18:52:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1d252fea34149c192704ac21e9f36c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73e47d4958ed4b11bf274678c27e11c7 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e608282883a4f848a19851e00c65f54 : SUCCESS in 6m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d723f80bfb5143f0902a655b5ec74ace : SUCCESS in 9m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b887772f1c94f7698cca1136bf537b1 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb121be8e7543f5920f26eb4585c66f : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/885e0ee51b7d47e381a2e6afe6d69b4e : SUCCESS in 28m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/760a5025f2e542a98860b1318caf2194 : SUCCESS in 34m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86aef327fb8b4deb957bc876e69c953d : SUCCESS in 1h 23m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cde6094a56d49fa9a7c042240fea474 : SUCCESS in 31m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97ca7fdbb5ea4cccab30fad299feb4ea : SUCCESS in 10m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/af31ee1f6c7e4c129a632c49d70fc1ec : SUCCESS in 16m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cc842c7f38694177a934f9649f59a52a : SUCCESS in 5m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5d0ce9f50c2e485696e020ffff608dce : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3e9560b4c95f4def86146da5f5d0c2a7 : SUCCESS in 4m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8eafc3aa9e2547ddb93d22873838b28d : SUCCESS in 5m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/80049336a0c6455dbf71d741534dc3a6 : FAILURE in 35m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2dd8f9df921a4f60bf8f6075bada4ef1 : SUCCESS in 7m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e5690b15fc3c495ab446851499df27a1 : SUCCESS in 1h 37m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6bdeeeee830b4a14be832544f731f783 : FAILURE in 1h 34m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ffed8f3f7bec4afe906be6ed9ee13012 : FAILURE in 11m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bfe65c23d5214fb59a5d22f02f871e68 : FAILURE in 17m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/085edc9fb93c4cafa81d245e8f019210 : SUCCESS in 22m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0f64df2c4d1b449484f2be1245de6002 : SUCCESS in 1h 04m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac8a7d80ad4448519abc762aca20836b : SUCCESS in 1h 37m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c283463349e040aaa4a860d2f60553ef : SUCCESS in 39m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/19ecad2f95c54e8fab6088c9f2926ff4 : SUCCESS in 1h 11m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5ba0bf0bd2b94f26a5b57ec613520f82 : FAILURE in 23m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/200ddc90d67c44c59a9ac0c3960688ed : SUCCESS in 1h 04m 15s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e7bd6ce32ab47ff9f624131c864787f : SUCCESS in 1h 10m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d683631b8e2e49eda6859cc1c0234885 : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":3},{"id":"328ef1c13a5b5f750b66b99d79008e4539480761","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-09 18:53:19.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-3/FC/ : UNSTABLE in 2m 1s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-3/iSCSI/ : FAILURE in 1h 39m 38s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"33ae66e643d40f4d7da13ea1a1c9adfb9f8e6ce2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-09 20:26:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950929 : SUCCESS in 45m 42s","accounts_in_message":[],"_revision_number":3},{"id":"7d00fd03f3eb7670cd7026f1379a42d811d70390","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-06-09 22:24:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950929 : FAILURE in 1h 19m 08s","accounts_in_message":[],"_revision_number":3},{"id":"08710309345550e6f3e04d0cefcb95f253f29a77","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-10 00:34:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a9649b3816174feebfc12f76cbcf1f2b : SUCCESS in 1h 13m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dcb87a555a1c4b97b6fe55bebea1a4cf : SUCCESS in 1h 23m 51s","accounts_in_message":[],"_revision_number":3},{"id":"99b79ff892cb45be13fc74c92ef7a24fdb71f368","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 01:15:53.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1048/.","accounts_in_message":[],"_revision_number":2},{"id":"0a466f790c3a41e22009fd0fa8e3d737712b16ad","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 01:24:54.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/924/.","accounts_in_message":[],"_revision_number":2},{"id":"f1f29510f713431f6fd44ead80df5fd1d1b40f39","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 01:28:59.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1049/.","accounts_in_message":[],"_revision_number":3},{"id":"6aca2ddd8bd1b757974dbd2f50ab26c92d06c35f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-10 01:37:56.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/925/.","accounts_in_message":[],"_revision_number":3},{"id":"9d4a5dd29ad41680982a18abf04dac5795c14863","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-10 01:41:48.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"fdca9b9ed42563d0e18caa904262409bbd5a4035","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-10 03:07:20.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"5f5ae65654fec0e230e28f21289500bbc29a6cc8","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-06-10 03:53:58.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cdec747e831bb4d9483fefa9bf15bde4f9bbdf58","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-11 09:27:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/3/20250611140317/NEC-V-FC : SUCCESS in 4h 20m 37s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/3/20250611140317/NEC-V-ISCSI : SUCCESS in 4h 24m 23s","accounts_in_message":[],"_revision_number":3},{"id":"48e6cb2f7b3cd734dec368056effbba03490b7d5","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-06-11 17:47:53.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a648ad3e2c9d0d61a1fcb5cf0d57926a7f28f009","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-06-12 03:42:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d977ec6554341ee110ee556f89db39f9691e3ebb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 04:15:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62a77106c4794bfe984c2fe77e06ced7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7b34706352941e788386f7b34280a25 : SUCCESS in 30m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0432753fa0b6463f8c7a7534441e2546 : SUCCESS in 20m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9b3784cd860742d68ba6615b2e98101f : SUCCESS in 18m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bdc869bda010424ea56296e0e27669bcca61709c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-12 05:29:27.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9735e71d76eb4bb2950ea01a8198cd8d\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/71b8d02d7da4414cad01f57fda7d66b2 : NODE_FAILURE Node request 200-0000066966 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a01f05899a6741b7a327ff3d8bfdfb24 : NODE_FAILURE Node request 200-0000066967 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5b32340d7f964e409afcdf6d893d1440 : SUCCESS in 41m 58s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6d22ceddec6846c88fa6431543bd6a85 : FAILURE in 1h 46m 27s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a8c48be07e0d4a1d80976a020bfb3c31 : SUCCESS in 46m 16s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7a9fe8444c0b480a8188ef5eb266d871 : FAILURE in 1h 09m 13s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/986322045fd74b358b378741cb660607 : SUCCESS in 48m 16s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/911aba97625646d2abc9d5df11e9e171 : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":3},{"id":"7414e67e0fde4d6235504808674ee7da5d97ad7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 06:23:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/212d577acd774141885bb8af83e2aca7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61120b3ccf214179ad70e318dbd60750 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/affbacca86c849109154412d1619a758 : SUCCESS in 14m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a0b90f1800d64630b95c3746d8d9bd0c : SUCCESS in 13m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d4586b74a424aa2b984a6b0041464f5 : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98ea2976a02a4bdb8410961ff44d5754 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/b9c8318e864f45e8810600f28d88ce43 : SUCCESS in 46m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b11ac3102ce4edbab696d877999dcba : SUCCESS in 49m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/657495a7f35b4d34bd3489dbcd849fbc : SUCCESS in 1h 36m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1372e2e7d6784bbfb190779c3c9d498b : SUCCESS in 38m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3091ca3281d448bf9ab382573985dc6f : SUCCESS in 15m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/648827cb1716410dbff8f5131ab28fce : SUCCESS in 24m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/af3c0fbf0e9941628df6c1572c01c578 : SUCCESS in 3m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/065d8f3eab78420496bbdda409f26b53 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dbb32ec6f6da46d2b58d53193168884b : SUCCESS in 5m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a2b44a94cd5e43a4995c25005d42653a : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/33e5cfe3fda3472abfec75986b219a36 : FAILURE in 34m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/08848f3c5fb340798a66019beb630c3f : SUCCESS in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f85dbf679b284c2cb0a5d9a5c6c986fd : SUCCESS in 1h 20m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6a0e6a6653ca41d7a91945285bcdcc49 : SUCCESS in 2h 29m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/212275bfe3ce4d0ba8ec87db8c7978aa : SUCCESS in 1h 58m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/72d40e6ee90a4bf4b89f8a50184bc087 : FAILURE in 19m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/09a7cad3c5444eafa4aabbb95848e76f : SUCCESS in 37m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cec24004fc954ddfa8b4a659b211ea7f : SUCCESS in 1h 10m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/caf1f411806d46dcbd25ae4216d5fb22 : SUCCESS in 1h 36m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d0c4fd67671e49f5a13b9a28d720dfa2 : SUCCESS in 45m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c798782d53c646218475d5620eee1491 : POST_FAILURE in 2h 02m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/75a63ab7c21043a7a9e2a408e6217609 : FAILURE in 20m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8596e3a74f36441ba77f70f5133f0512 : SUCCESS in 1h 51m 20s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01948be2e4104a29ab645e60c09c3f3b : SUCCESS in 1h 23m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d26ebe5bdb8499c8b96d90c9f97b2cb : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":3},{"id":"036c091d29b8f6c861b08133612d9a70928081d5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-12 06:26:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e675e9a19f5d43ad92cd82a2c4e984c2 : SUCCESS in 1h 15m 20s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0469f50d2a2c459f98ed2c85ab20cb4f : RETRY_LIMIT in 2h 41m 41s","accounts_in_message":[],"_revision_number":3},{"id":"7b1b8073e9500bb216a07973cc013a8c4ab06d4b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-12 10:45:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/3/20250612094921/NEC-FC : SUCCESS in 6h 37m 55s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/3/20250612094921/NEC-ISCSI : SUCCESS in 9h 55m 38s","accounts_in_message":[],"_revision_number":3},{"id":"c9e00e1d1ceeb2a1f39a812b45aa344e85e261c5","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-06-12 12:30:51.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a99e79ee613ac1a2fe519d797369e968216ced23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 12:59:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acaf2c7b548b44c7b588e29d36494093\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b02819b696e544ebb9a8d3174d94c086 : SUCCESS in 16m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a1adfce7ad9743859cee0a926964c3a8 : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/efd3f5742cb444d9824d328eadfd198a : SUCCESS in 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c895ba26c221549e563129d7103bc2af0644625b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 15:05:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e8b7a4b8e03475ca7a485598cea7332\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad8b1b07db94ac1bba1ce9be0b4f9d5 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8bde445974b42c798babb3b19f85871 : SUCCESS in 10m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b3f77570abd44d4b6db1cc6c8eeb2cb : SUCCESS in 6m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30b8d731d1aa4c0bbc74c4da6597de30 : SUCCESS in 17m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/243f66f1b1514b3ebc02ee030b87adda : SUCCESS in 16m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/da5ced35626049dd99ce3548d771b8f4 : SUCCESS in 58m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e2615458b69544f2b17f429c9763af07 : SUCCESS in 1h 06m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18c1cbf822684e2f9ee1f485ff6b0fa5 : SUCCESS in 1h 20m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd2dec99a01a419f8b62da1f282b894d : SUCCESS in 1h 00m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c16b72ca57f46d4817f977bc7904558 : SUCCESS in 17m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1bb3cb89c7a84cc6b5c31c84e4b39f23 : SUCCESS in 26m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d6c606a739c747f1aca361e738650458 : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f0692e118a14417faff8a7ad63fcb052 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a13a3f981b574506987339a33278eaf5 : SUCCESS in 7m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6299b4c9d4634df7aa28ff87422fdb01 : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f650a3b485b54c369d64fee348c0da63 : FAILURE in 16m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/426b5b3f9de5448cb758b791f87f156f : SUCCESS in 7m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/754e46315d464064a2ed2d286eea2798 : SUCCESS in 1h 59m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7c4a75abdfe0433ab1e27755a0229911 : SUCCESS in 2h 31m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8a23c844043e42a6a37aba907c0ef4d1 : SUCCESS in 1h 55m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/876d6b6d05e643c0b25605dc6d5733c6 : FAILURE in 13m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/addc88c50f3e4f7f89ae5ab6bef0a2fb : SUCCESS in 21m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e6f4916ea80f4c0696358fb13dc03999 : SUCCESS in 1h 00m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5366683927eb489a8f32a73fe176a6d6 : SUCCESS in 1h 33m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/207f0ae48e05413d87ebe9cc17cb8d9d : SUCCESS in 37m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2ff593dde1264cb9bf5f17743b18198c : SUCCESS in 57m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/48cee42f2bc34a3c8e5d813f8de80d4f : FAILURE in 20m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0078257f0c104a5e81efa3b224caa69c : SUCCESS in 1h 57m 59s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9b82a6771ac42419752dcfe228650af : SUCCESS in 1h 23m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0f2d5fc4398432bbcf4409a29a28387 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":3},{"id":"4c133419e779c230acff6a1f3291cb4039f337a8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-12 23:59:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/aa2eda7f406b41528ac6196827eb8f03 : SUCCESS in 1h 17m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/41aba0a8715d432690264c99041d21c1 : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":3},{"id":"da09406bf5d362893ab571ee0c4232a184535c50","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-06-24 11:51:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0bd6c3b734fb6d32b0cb1e3b838c08faf14731b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-08 11:11:36.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"75b983c901ecacb50c7fce19e31fc60093237965","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-08 11:12:29.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6ec4ac6f8bc441def84995634753018f21f54de1","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-08 11:13:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"91cd5ac32cef21a1a6afd4c85dce90441570b3eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 11:39:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4f5abfff62a4dd2aa39c78b628c4e11\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd62292c2a3d414ba12ecd8f889fdd34 : FAILURE in 16m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6985fc44d88549c39ee450b0c32681e3 : FAILURE in 21m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ca7a067d6a04e27b236d5f9495c7f86 : FAILURE in 26m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8396ad9d9d9dda319a55c7e81e1865ca662aace4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 13:45:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/012b4e8d4ed64b13a928867f5bcdeaa5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f8df4aca3241c88d71085b35d8a566 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/965db47c3cbb483a9597180cd358fef7 : FAILURE in 10m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b8981cff80845d8b78badcb018781b1 : FAILURE in 12m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f60f41ec30404ae39ff60f469a30fa9b : FAILURE in 14m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea2569772dc463391c5921c3cef128c : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/45e1867c33e343b7a059ffc26fe5cc3a : SUCCESS in 45m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/805bc55723f74a84bc524512998c54a1 : SUCCESS in 46m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d7159df54cc43e7bf09d0ecc198c660 : SUCCESS in 1h 24m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/871eb6f7602646be82ae5cacfe1cfe87 : SUCCESS in 32m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9b84a214a8844b4957e32dbd1208d14 : SUCCESS in 13m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/93c30961216f4c2a889c9894d2016f2b : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cc90ca1c3cd94930b77d608476bf3ddb : SUCCESS in 4m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b302704671194f35b072b01164315872 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/510f110fba3d4f82ae1213c7b9ea1110 : SUCCESS in 5m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dfdab935e7334e0b983e4d0c399a6a3a : SUCCESS in 5m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0aaa275d5c09483994d9657c9888edd7 : FAILURE in 31m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7c2808a724645e8af9806f02526d79e : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/267c8df6628040068f1631107567160d : SUCCESS in 1h 50m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/11536a1e9416410e9ea2425b7ce876b7 : SUCCESS in 2h 32m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7fdc4cdef6b34f38a746422dbfbc7b0d : SUCCESS in 1h 55m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2e74ccdd00a4470d91f66168e026850d : FAILURE in 9m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f1022abca9c84a1d9d9345564cc16207 : SUCCESS in 37m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/853c4d43e2634ae9bf8ee96258467db7 : SUCCESS in 1h 03m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2211cec2159f4cbeb1d1effbb1099058 : SUCCESS in 1h 44m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dedd9cb3940c4858a38969c22fb69c45 : SUCCESS in 42m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8d0c6ab9acb94c6981e757d0c4c9c786 : SUCCESS in 1h 23m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/52958cd2fade4540a49cdeb0c3fa9b4c : FAILURE in 9m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/361f1756af984ee1b4a739846d1ff070 : SUCCESS in 1h 50m 03s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c60bec4b32ed424895de461349a59608 : SUCCESS in 1h 45m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07ff7f11693041b58703e56c2adb5ee6 : SUCCESS in 56m 12s","accounts_in_message":[],"_revision_number":4},{"id":"a9f3476150a29f17b0a6430986fa5eb38ce21c74","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-08 13:54:04.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d7fc0295af524689a5322ee05df3dbaf\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/97ffa6d1740d4092a1d61863b6972bb6 : NODE_FAILURE Node request 200-0000072184 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ecee576b697f498a9559fb4058ec58e4 : NODE_FAILURE Node request 200-0000072185 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5f79bb99d2074f61bb1718e728fa2b17 : RETRY_LIMIT Host unreachable in 49m 09s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a6b8a099fd4f4f0fa6eae08804a62149 : RETRY_LIMIT in 26s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6eaee81f83b54012a50ec2cbb2a60b20 : RETRY_LIMIT Host unreachable in 46m 09s","accounts_in_message":[],"_revision_number":4},{"id":"0be81bf3e37373e1b54960b2c7f6f02f11170a11","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-08 14:56:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950929 : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":4},{"id":"ac0e4cb057a3e4f417834e30552108997852311b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-08 16:34:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1190/.","accounts_in_message":[],"_revision_number":4},{"id":"ae3fae726c9b79e2a0498a6e1564c52b63847309","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-08 16:37:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950929 : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":4},{"id":"4ab920a3ae6e5a0239d9a7ef9eda564049e1feb6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-08 17:27:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1320/.","accounts_in_message":[],"_revision_number":4},{"id":"5e1933ac28555d79ca2c8c10123e8045a8caf360","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-09 01:12:15.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"bde10772d7d5ff1fe5a31e048f5f752e8bd2373a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-10 14:59:48.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1f92bd252736c9012d378c1925d66061fb92674a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-10 15:00:35.000000000","message":"Patch Set 5:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e2ff78933ac11f01b852c2bc3092d55c957f10c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-10 15:20:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3909aafdb72882bd13a94bdde3821156d310df7a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-10 15:21:26.000000000","message":"Patch Set 6:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":6},{"id":"dd7acdcc6602980913a1c3879aa3329cce0c6927","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-07-10 15:27:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ea416c2e332813c290a4e4c2ce46843061116d45","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-10 16:13:20.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950929 : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":6},{"id":"827b4abf869e033e6dd7b81f7e4b3b3e50c1fe68","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-10 16:58:55.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950929 : SUCCESS in 1h 00m 55s","accounts_in_message":[],"_revision_number":6},{"id":"ce8bdcc2296b4951a3330c56392c0a44bbd32bdf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-10 18:01:01.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4f323ed7d06f4ec0968cf49deb353f41\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c88da17c66f64c46a15fdea4c05c8cfd : NODE_FAILURE Node request 200-0000073555 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/53175efec18b42f1b533870d010e1348 : NODE_FAILURE Node request 200-0000073556 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1126f360590a43088f064aebd9f8f860 : RETRY_LIMIT Host unreachable in 57m 15s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b186eac37595426197044a4763999eff : NODE_FAILURE Node request 200-0000073576 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6430921728d4120bec4f991b5010470 : RETRY_LIMIT Host unreachable in 52m 18s","accounts_in_message":[],"_revision_number":6},{"id":"ffdec686c912f5c467d71c4d75a56dc99814616c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 18:24:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/811253c0817140adb16a6e9948f4bf60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5556f1269cce4f608a35fdbef259edf0 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7dacf1407cb947eea2d9ed94c2fbfacb : SUCCESS in 7m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55eaa099dbc24b42bb720d940839981b : SUCCESS in 6m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7ab20b5bd3246ab8ef4a1d48e23ac45 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/611e69c10f3745a1b067dedfac430427 : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/736c79fe7a9345df8e613f9d26e0ede4 : SUCCESS in 29m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2c8530a3042a4bebabaa19fcc8affe1b : TIMED_OUT in 3h 02m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6d7e924000c4e2b84503559e837fbd7 : SUCCESS in 1h 21m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae8972cc8f6d4bacbaba2b6322788d7e : SUCCESS in 38m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6222f4df42dd4a6c85c1517e9a9c1d59 : SUCCESS in 8m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8bbdb3cb354c482090a7a37edd3005cd : SUCCESS in 22m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/97f235ae08934505ae2b3f821a8b090e : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bfc35ab605ae450f8c3ae778df8317a1 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/18de4648cb4349b48848d142401b62dd : SUCCESS in 4m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8d4a780e570740d4a872867d98f704e5 : SUCCESS in 9m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1522f7ca17db46b7a97eab644c2655ff : FAILURE in 36m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/096e35fe53434c7aad1322b049f0599b : SUCCESS in 7m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eb32bfecc58d40f5a92e9fb055d8ee66 : SUCCESS in 2h 04m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/363a5df2e84d40979b8444bdc27a3a03 : FAILURE in 2h 13m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6edb3e7d53d94f77ac55382d24198bc5 : SUCCESS in 1h 39m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/81740890530045c4bd8d7cc769465cf6 : FAILURE in 20m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e48d5fd8019c461195f067cd02d03edb : SUCCESS in 25m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5e47933cb3a5465d88927adaeda7fa17 : SUCCESS in 1h 17m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d7fec38fd7e042298dd5ea803579fff6 : SUCCESS in 1h 38m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7e048b27d67c4b28bb2308a4384090e2 : SUCCESS in 37m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7c3bd0c0ff9841efb2cb1a95f4d0bfa7 : SUCCESS in 1h 25m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5591f8d6c1b648ba97fdb4643d0930e5 : FAILURE in 24m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8cda4df528d749c3901dade4eb1450cf : SUCCESS in 1h 19m 59s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/821280241e5141e2b2a439e2e57bfff2 : SUCCESS in 1h 29m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47fc0095526b49bf8a3634b442afadfa : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":6},{"id":"5c0198d2bc28372a32694463bcb2fc879386f5ce","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-10 18:47:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1219/.","accounts_in_message":[],"_revision_number":6},{"id":"72f50e5be023ceeff81d1a70ab209f1faf541e68","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-10 19:11:44.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"2b7694e477396c142c4aabfeb402e35575c87fda","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-10 19:17:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1349/.","accounts_in_message":[],"_revision_number":6},{"id":"ab6df418bfce578e03cbdd83db71ad69b1f29268","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-10 19:45:54.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"50be1e906c443978e92a379cda8ead93600948b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:40:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdd6fb420d82469887aae204274cf279\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5380d973d47749ba8c9cd389e84fb8a6 : SUCCESS in 15m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ecd42ebb3ce64aa587a9b4e288262da8 : SUCCESS in 22m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/06246d3e1bea4116a9105822872a33d1 : SUCCESS in 26m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bdb4cdb45e15df95f332f819b62f6bc3b75dcf82","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-11 04:12:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7f74b497d1f25fccc8e7f683cef68ea97b96a72e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-11 04:13:40.000000000","message":"Patch Set 6:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":6},{"id":"87e12827881a2fe123097c232d271026f89939b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 04:48:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/659504a1079b4b1cb34cf9b526e4dc87\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc1e3cd495f247f3992cdffe6c721890 : FAILURE in 29m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8206639dfd1647f8bfecdb62c36aa6af : FAILURE in 29m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/217cc47943034693a94d2bf9bbe7540c : FAILURE in 33m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c24480b1a2d6cf9d0aa7651f3f9fd2c0546264a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 06:36:41.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c0dbbed1e214e6b9eea4dde02b7bbb9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e35cec1efff437485162709f5e929db : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62a7bbe91f7b45d0839962ffb3545179 : SUCCESS in 8m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac6509f95feb4b71a2365cde220cf24e : SUCCESS in 8m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b7e6b43f01a4572a4c6f9646c3a7f44 : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2870cc4566488fa9a62bfd5102ac34 : SUCCESS in 10m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/72ac2e099b6c49f8abb33236b566d4fc : SUCCESS in 29m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a5aa101e9f34bb0a6860ce5827dd61b : SUCCESS in 55m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ee03610af634db9aedfc062a3b11b64 : SUCCESS in 1h 30m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77135c48a281425ea4452bd25e392cd8 : SUCCESS in 1h 03m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79aa89afde234f9099c160cb7f8001dc : SUCCESS in 8m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a7386071d03b47059371a26d5f5180ff : SUCCESS in 22m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1ecbe4ad32194285aad816f01d805a0b : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/67bc1e9ffb4c4a0eaa78244ae9aa2090 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d63de19c95e64f10bc45e9087a8bb94d : SUCCESS in 4m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/543f93049c23450b8129102fbe4dffda : SUCCESS in 9m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/05f34817f1964eb7889fa30f4f023cb7 : FAILURE in 30m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1e646016a71144b8a79f9f69747ffe33 : SUCCESS in 7m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6381c33699144a629d65b29e9b73da40 : FAILURE in 2h 10m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/42769df1575b47cb9e452066296f22bb : FAILURE in 2h 21m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a365ba2dc7dd47489048da7bdee701df : SUCCESS in 1h 57m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/35c5a65a9c0046cda517ee4b83970538 : FAILURE in 9m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fac5e4a3fb434a94b98229603adc9ecf : SUCCESS in 33m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/430e38a51cac4e758063738d60883ace : SUCCESS in 1h 02m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e644a1bb81be4496906f60124892306e : SUCCESS in 1h 23m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fb7483f41b4a4449be697aeb7e907701 : SUCCESS in 36m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/45e9096902134b589437c4292b99240b : SUCCESS in 1h 14m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ba326d8389f643e4a403c31f259a7a60 : FAILURE in 16m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c70c6683af3d415e8f7850d12117bd5d : SUCCESS in 1h 34m 24s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a76fa0eed45d4e26b81b147d22a8bda9 : SUCCESS in 1h 11m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acdabfadf4dc4545a68d483583bedede : SUCCESS in 49m 19s","accounts_in_message":[],"_revision_number":6},{"id":"38ae5b14f90874e448d0615061f8b8c838ea2359","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-11 06:48:21.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/396027c20f9a4e6184a56c83248cef73\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f3bc14a368ab4d3aae75e1d05aea49f8 : NODE_FAILURE Node request 200-0000073621 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0ca253783b984abe8cd19ed4d9af793f : NODE_FAILURE Node request 200-0000073622 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/64394c4a60f54ec5a86d29b6a21f2625 : RETRY_LIMIT Host unreachable in 52m 03s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4e1f719d1a254d99a415c88efeb03143 : RETRY_LIMIT Host unreachable in 48m 02s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/61695bdadf864ff3b299c92d4692ea70 : RETRY_LIMIT Host unreachable in 57m 08s","accounts_in_message":[],"_revision_number":6},{"id":"4c04cc390b84e4f133fef45b19e0f78565b38a0b","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-11 08:40:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e617cb999f146e4d98df2e7405b868e69e20363e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-11 08:41:33.000000000","message":"Patch Set 6:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":6},{"id":"37fc10efa1556e2cc2f6a6d3bff5346e13a9e279","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-11 09:26:49.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6d3eb02a92094179b23dbd872cd5a438\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/77229cc1c0f846838882903119caf3e9 : NODE_FAILURE Node request 200-0000073655 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b474fc99caad46dab526055075befc3c : NODE_FAILURE Node request 200-0000073656 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a88e7ac59d4c4f16bac870a3b1b0fcfd : FAILURE in 44m 54s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9bc848702d3847bda40b541f249541d9 : SUCCESS in 45m 31s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d3015f3b871e4711b078deda8469213a : SUCCESS in 43m 11s","accounts_in_message":[],"_revision_number":6},{"id":"48c4a07d3fcab9866b93d81e2c96b04dc7b87626","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 10:40:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abadb2ba2c6246e1a89246a121e608c0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/798cc4bc3adf45f4a57832853881f6cd : SUCCESS in 20m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c21fc6713cf44f16a7a6d3053213f17d : SUCCESS in 23m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/39c7aebe414e4c95a9e7869f3e3bf116 : SUCCESS in 29m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0a5ff70e0c783640a059c1237f05325cc6e82b64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 11:25:47.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fce75c617d7d4e8da8c98122252d23b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d9aa19dcac431dbd99e530fc252edd : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68b3f2c55c8a419d940455d133bea51a : SUCCESS in 9m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54984bb31d604e728b50838101761083 : SUCCESS in 11m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f275052936a74a769f7d456b0dc29d8f : SUCCESS in 13m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/682ee425f16c4f29a11b8416fb03e541 : SUCCESS in 14m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/cb7aa28d81754d64af54c5e50b223201 : SUCCESS in 41m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4bf9a407cb4247e3b444d4d269c7d999 : SUCCESS in 59m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/80efb9468a0f4ea6a9b9a56add26ba83 : SUCCESS in 1h 09m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/198627c1236a4611ad51902a4a02fd34 : SUCCESS in 58m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17fc0aa985124ee687e80d84413c31ac : SUCCESS in 15m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b756ddeeffa44341bccb981efab7ccd8 : TIMED_OUT in 40m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/95cb6502aedf467f955b26f9d80f2afa : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e7486217011a452ca861b7e074529733 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d07c8d9c5f134f09b3856f1e98836551 : SUCCESS in 4m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bb767b376c3048adac9644524077fc1e : SUCCESS in 7m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1b3b5760346741febbcf5284d59727a2 : FAILURE in 36m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88aaae83c77d4221be54bc44af550872 : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc694a1690dc465484bec1a99c335c27 : FAILURE in 1h 31m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/89f60b59ab3743cf94b1cc1da6521f34 : FAILURE in 2h 39m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f044592abf004ccdbc34992edabec39b : SUCCESS in 2h 06m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/31cecb8e92e442e68559c9f1d364697b : FAILURE in 23m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7b2228637ae54d36b6b6f3deb2ae526e : SUCCESS in 23m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e80230e1cd8c49d898a85fa04e558892 : SUCCESS in 1h 01m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5874b1cbebd544c798da50b83fc8a337 : SUCCESS in 2h 17m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/845274ffa2d3406b9b1c9dab3ad5079e : SUCCESS in 37m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8709e7c9e98d4c09924407f2aea38432 : SUCCESS in 1h 37m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/63b7195e0ab14b46be0ea309f11446c7 : FAILURE in 24m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/30fbffd598be43d98e3ec899353ffd1c : SUCCESS in 1h 36m 02s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e9d8d41d4734c188e4d79887f91b013 : SUCCESS in 1h 28m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50e4273ccae047e4bbc0e559982f57fa : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":6},{"id":"d8cff336ca638d713b9066486aedf992de288c36","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-07-13 12:09:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9048e78fee474ee08e9c7d5b2cfb847ba3449391","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-13 12:10:49.000000000","message":"Patch Set 6:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":6},{"id":"173d2d4592404907194ce711a4a2b3753c0a9bc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-13 12:41:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/195425bc951a4f4fa3c0027ca78a33ca\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a55e12d00ec142b592e90a695fdf6493 : SUCCESS in 20m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/660c1ff39b9243ea98f97c93952d70b4 : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3d882ef5e9e04b1085a1a1ee9eeedd60 : SUCCESS in 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d507d7a0340a087fe8edf741c79ed322c13bd3b6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-13 13:04:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/56f9cb61507d473b8885977dfd5bd73c\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d6586f9ad2e740b7b1f8c5b7430c4004 : NODE_FAILURE Node request 200-0000073751 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a1b79d81a9374079856eb3e9739152a4 : NODE_FAILURE Node request 200-0000073752 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1ffad1b9451049079910cc7da104d17d : SUCCESS in 41m 09s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a36f28b14c434163b90436693e7c1367 : SUCCESS in 54m 18s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ae3d6013f2ba4acbbc1953d2b77de0ae : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":6},{"id":"c33aa6c0f9e8e1a3ec3b48cbed5236e0853fda40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-13 15:19:08.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4cb780fc93842c7a80c2490c128b368\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4937a2f2e36e4200a680d0c490a3014f : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/deb11124ad874e438e05a6e9f7378896 : SUCCESS in 11m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02aac8c819d645afacaa2baefe9b2860 : SUCCESS in 11m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2eec354697a94fb8bc260ec832fa9853 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d80e6ffa154397a48e6186e3787318 : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/b5a154e3c9534023bcdc8c87d4126359 : SUCCESS in 47m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/06727afd13384a8dbb4568c836e7268a : SUCCESS in 55m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00091e9180754d57ae1c0476306a674e : SUCCESS in 1h 26m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7dc7ff6e2574ae1977209fb6ddb53c7 : SUCCESS in 50m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7791cd80825b4ecb9c90f67f3c59db99 : SUCCESS in 16m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/de345f1bd2aa4ac1aff4c5ff351d41b0 : SUCCESS in 20m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/02ff0983e5cb4b43a1b3d2d28967fb91 : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a3dbed183c6a42eb94f31f6288488fb8 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/92133db052db46cf9938741415acec93 : SUCCESS in 5m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/994de39611ba43e5865acb5aeafcef03 : SUCCESS in 8m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fa35b7caed2049b39668babd921fb73d : FAILURE in 28m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/57abd5399a2d468191cee3abb64d2b03 : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4617763c99e646d4853bbbf50c6294c7 : FAILURE in 1h 34m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c094b52aafe940dc9c0b415b2caf4313 : FAILURE in 3h 04m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/022e22be23c448ceb8f29d1e510eebb0 : SUCCESS in 1h 26m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a8648ff7fac444d1b1f5bd4a17d06410 : FAILURE in 14m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2e6440b6cf1e497e9668fe8aa62d4749 : SUCCESS in 31m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/99a37b0960ef44218ad9740b75be17dc : SUCCESS in 57m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e46f47c53df4491ac4c38e899168ab5 : SUCCESS in 1h 52m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/044a035c94874b9584b69f7cf0521997 : SUCCESS in 34m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d96fa25e08854650a751b56464480e6f : SUCCESS in 1h 37m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3e71d8e1d056491cb9c9cc0ff99f9dfd : FAILURE in 21m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd2a240191894ced9c74af0126b2cf81 : SUCCESS in 1h 12m 41s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e7b4c4e79c2462ebb2e11e0c2bc3368 : SUCCESS in 1h 13m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d72d3b5a4a4048cea62deda6941fbce3 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":6},{"id":"fe6f43976f4dc2259ad4fdd7cfb86d424cc0044c","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-16 11:13:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1593b5c8c6f20e23ce8768b4687fd9793355822f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-16 11:14:37.000000000","message":"Patch Set 6:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":6},{"id":"486765299004212b1dc26209349ef64d0dc1623b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-16 12:09:58.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b51dd9f3a38d4b18869a4ad74dac0ddf\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/082e5740e0464730a881334c0ee7defe : NODE_FAILURE Node request 200-0000073875 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/11ec2fd840844f05955c3f3c935f32fe : NODE_FAILURE Node request 200-0000073876 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/39bf3b8674594ade82766fac6ee252d9 : SUCCESS in 55m 39s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b869dade58e42e4a7f564486a9d4197 : FAILURE in 51m 55s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ce793bcff16d40beac9cb147240aab76 : SUCCESS in 47m 15s","accounts_in_message":[],"_revision_number":6},{"id":"637cc4810d7508ed714f4fda3cbe4b36ea6fbbbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 12:38:17.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d15c91b6f404c1b8c44eebe4dfd2054\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/61db26c75ba84d26b26f1da65983bfc7 : SUCCESS in 26m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b4de5e3e057417e96852b857a614dd2 : FAILURE in 28m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0faf633666a54d1c87187bd8c0fa40e3 : FAILURE in 33m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4bc89c03fd461db027273ddc341244e425bcca58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 14:29:50.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fc74e6fdbf0430d9f07773d5b26eec8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3cd8f9beaf84c218b95c3087a527bb4 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d2685468bdf4a79b15bbb3aee1f5065 : SUCCESS in 10m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e24a0cea4be438d98fe29ca250d61a0 : SUCCESS in 8m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd450911e57a4cfeb6aeef54fecae26e : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f21bca9fa757467ca6c6ee8b049a04ed : SUCCESS in 14m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/f3c1bfa80d1b4cae999599514deeb4e8 : SUCCESS in 50m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/491836ede9254e258abc3a72dc2d7516 : TIMED_OUT in 3h 02m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd96e5f22dce4ff1847be9ef16ff1b4b : SUCCESS in 1h 24m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cbe569d4dfa47a9ad8918095db83e98 : SUCCESS in 52m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc1e12ebce3f43e1a927488a25083bc6 : SUCCESS in 13m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d94fb3733b254fee89ac12d3e8fcc8d1 : SUCCESS in 23m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9bb97783063e4c268f73a50d4f67999a : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a58fd6cc4422460ca52cb3967f904726 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc744698e4cd48cb820f5c3b5fd918b2 : SUCCESS in 5m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ddb08d203f6d4e5a8a757ba415331b47 : SUCCESS in 8m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f3b94a321da2421cabf097426bfc9f1f : FAILURE in 28m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/04aba49c498b4bb9862bc145950e92c6 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0ddedb5851ce4ae7a3652223a60693cc : SUCCESS in 1h 51m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8af2121574894bb194c8949cd8b83be7 : TIMED_OUT in 3h 07m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ed2ee36beab942db9100ea4bde3d15f6 : SUCCESS in 1h 40m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/11e1b51a002648dcb8bd05b8ccf5cc1f : FAILURE in 9m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a5196e95ccdd4ad48667d1a607c563e7 : SUCCESS in 30m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/484218f110ca4759b6a4aa1e749ec70b : SUCCESS in 1h 06m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/928c2aa4f3624864b4849bad40740c6e : SUCCESS in 1h 27m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/44993f878ab7428888cd5748e234d64f : SUCCESS in 32m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8e9c0eafe3344730bef07e0be58d53ed : POST_FAILURE in 1h 59m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/677859808bc14c5e8bd0daec9e627126 : FAILURE in 19m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/34a467912297412c8b8867c9201e1575 : SUCCESS in 1h 42m 51s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ba91da9a853455e80e5e612aee1e90d : SUCCESS in 1h 24m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7402175f42164e9ba786d59a9cc29d3f : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":6},{"id":"a435bfebe2280ccc9ba39e3f9bd2c6c0835e377e","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-16 15:02:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"113f84ded732ce981bcdef9e2e55cec01c8f2da5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-16 15:03:25.000000000","message":"Patch Set 6:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":6},{"id":"885d045e2e632276c33093f16b1cff0cc00e6661","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-16 15:54:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3213dee7a3a4469187051edade598ada\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/deffccb1e8344385a1296e7ff48cc8d8 : NODE_FAILURE Node request 200-0000073924 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3a176a26df6042e2b08d59d1e0922ca6 : NODE_FAILURE Node request 200-0000073925 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/18eb81779e444199b53f1b5f056682f5 : SUCCESS in 43m 25s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d6b73c7e598b4c7ca5d289c04292bca2 : SUCCESS in 51m 13s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eac3872576ff4ecdbca3e8c079d87590 : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":6},{"id":"9991fee2a36227229dff8ea6fea7dd1d5d61e10b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 16:37:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed17ba8da3bf40a6b20336b3fe1689a7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6eb98076e4fc4de68f1467fa93963f22 : FAILURE in 29m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d8ce299b53184b339ce94a2764d5ab7c : FAILURE in 31m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/148852f3e6754533bc200a01646110ef : SUCCESS in 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8f974e5d079b96b59c884438ce8a7c7797b5abd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 18:22:43.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/862b6bd7fc764f8bb28f8a216a5912bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e4010bcd7d742e4b23c6b192e550551 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e75ff3b40dfd44478acf691351eab6f0 : SUCCESS in 7m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5532a16def674fd183f214f8439d729c : SUCCESS in 10m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f14f2688511b4b0bbf174cbe94143f01 : SUCCESS in 14m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a8b04e86184127b5545fb90b00fcca : SUCCESS in 7m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/310fe22b5e174a3fbe5fa8e51c794e19 : SUCCESS in 44m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c485b4928264aef841be24420ee1c2c : TIMED_OUT in 3h 03m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d2076b6269a4f98b6217e327933be8d : SUCCESS in 1h 21m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80fbee42412349d8b88a65ec7228fd8f : SUCCESS in 51m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/beafa6f2c8c647489deabef1f81191fd : SUCCESS in 12m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f5baf99e19be45838acb47bfb20f804b : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/324a54d4d1ca4a42b50392da49692b6d : SUCCESS in 4m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f6f4f07262b6462a9e9d76a172ac8790 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf9d6a8309df4ac8931f3e4c97bc38b6 : SUCCESS in 5m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/55a6dc789dbe481a81aefd114b63879d : SUCCESS in 8m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8b77d953ee17483580e0bbf3bcb65f65 : FAILURE in 26m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/522a3a1c3dd04faea0b1f0ea488edbd2 : SUCCESS in 3m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc1cf91f3cdf43efbf5df1803bbf415e : SUCCESS in 1h 06m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cec7ee011a2141419cd9021c1904789a : FAILURE in 2h 24m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e1b5cbaac5d34775a1285dba3d3f652d : SUCCESS in 1h 55m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7b5722881e9643b6be020adbf91e486e : FAILURE in 24m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/231c93ea00b84ac9abb2a143a4838ea9 : SUCCESS in 27m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4ecae81f01124d078e8ad4cdbb8f7526 : SUCCESS in 1h 11m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/144ff58563604d65aad5a659cf89d79d : SUCCESS in 1h 35m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/305067d09b864e3abb5fa6c288daa26d : SUCCESS in 33m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a81b975604ac495e8fd5d4fb58677720 : POST_FAILURE in 1h 59m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bb91d43801b3441c8876868cc5e889ca : FAILURE in 15m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5423e6c367e641cbb3d0f52435481f3d : SUCCESS in 1h 29m 51s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69ddd3049a70440ea03e2da52eae1a06 : SUCCESS in 1h 18m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b218223a39643c29608a9bf4437f43b : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":6},{"id":"1656f352bf1fd667b4ab812c45550c48aca168e7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-17 20:47:11.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/6/20250717202025/NEC-ISCSI : SUCCESS in 5h 47m 10s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/6/20250717202025/NEC-FC : SUCCESS in 9h 26m 43s","accounts_in_message":[],"_revision_number":6},{"id":"65f464f6a5eb158ce9a9019b8726243450735f06","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-07-18 02:13:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7edee2bc3ac82a9b749fe7e1d0af8a7329e8b85c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-18 03:10:21.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/90fbb877061e4879a67af16314b71b98\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5f7318f988c64d7faca96999d738ee89 : NODE_FAILURE Node request 200-0000074553 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9e763a71d1044c0ca8c14ccf3c6129bc : NODE_FAILURE Node request 200-0000074554 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/477187a2992946f9b8e23a77fca5b01b : POST_FAILURE in 51m 09s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b582d928e484c679c5b9038048c0b29 : POST_FAILURE in 56m 12s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/40105719c38a4b2c8bc57abd8c76c2a6 : POST_FAILURE in 47m 00s","accounts_in_message":[],"_revision_number":6},{"id":"f10eedd507584ae0fce81d5e932dacf357015b2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 03:44:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c0ed6c7a32f4ed6b9b33de8e24cc392\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/327f6fb8fd9942d3902d814ddf99885b : SUCCESS in 23m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2107fb29edc446afb636cc2c71ca92b8 : SUCCESS in 23m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/02311433422947fd8ce83d8bf68a867c : SUCCESS in 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"92dbf31c4dcdb5753c6d8943818ebfb443e4716b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 04:40:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b1177f8dd09400bae5989c7c4b8453e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a31798efeac047b79182fdf71d22f8db : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e598364d2254a03a71bd91794c639ac : SUCCESS in 10m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22a316e98acf4673b4319c848d78d8bb : SUCCESS in 10m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80ffebb664804f0a8fd452b2f81cf262 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc6eae26ec64bc39089e3cfcdee3da9 : SUCCESS in 16m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/f2c9bfe0d49a4a118068bce605319f39 : SUCCESS in 56m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de8f0343751949979ad2089dfcc59490 : SUCCESS in 39m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b821ae6afd454f1aad2575b77d4338e7 : SUCCESS in 1h 32m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cabf3bf719a4758a4057a4adf19cfb0 : SUCCESS in 1h 03m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05710559f14147259d1d28883f369127 : SUCCESS in 16m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7be88c8ad74b43f0ab6e2357c622526f : SUCCESS in 21m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/75be665c18fe409a88a97965f7d510a6 : SUCCESS in 5m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93ccd476dccb4c92b5ebc21539d5ff53 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/34626cfbdd54450aa5f58054faa9a157 : SUCCESS in 6m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b630c20ca3b84c3cbce21d04aaed524b : SUCCESS in 9m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/78e9d6ebb9174c99a56178857715bf5f : FAILURE in 29m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fcdbb92dc7a0470a8cc99dd339b3fefe : SUCCESS in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ed81702f63fa440e9c83f096f8845f61 : SUCCESS in 1h 53m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b9c8147ceab4c85a6e7286b46a1964c : FAILURE in 2h 19m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/30e50a703fdd41888ea7914da0fcc40f : SUCCESS in 2h 11m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fb40ad82413145408e6c31ce308695fd : FAILURE in 17m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c5a2485edaf54b9285695b43ecc862b1 : SUCCESS in 32m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f86d78c1f1994a86876fb09a4eedd497 : SUCCESS in 1h 06m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0a77fb33c6784ab7979c7253acf5cb17 : SUCCESS in 1h 35m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ec2dec056e8a4fb98168676d4950de39 : SUCCESS in 39m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/121ccea61809460ba4d4168847f7505d : SUCCESS in 1h 27m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7daee76d5a8f49b6b4128888be9f82fa : FAILURE in 23m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0ddef652128b44f6b4a1a719149a73e8 : SUCCESS in 1h 26m 27s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c39794d5c6b49abb9193c63081e3cdf : SUCCESS in 1h 39m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf514ea5625b4c798581ba828c58a927 : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":6},{"id":"948a9410f7e64e88ce55bfc787e7bf17aa51758b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-18 07:14:56.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/6/20250718104330/NEC-V-ISCSI : SUCCESS in 5h 10m 38s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/6/20250718104330/NEC-V-FC : SUCCESS in 5h 31m 23s","accounts_in_message":[],"_revision_number":6},{"id":"939170f7da00754f7a73f5dbfead3cc5a4ceb403","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-18 07:52:39.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/427961adc8b84a5ea2994b3ea721a3ad : SUCCESS in 46m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4f83b7fad41b43a89415ea5c4d3ae29c : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":6},{"id":"c9bd0566b70c459b0c268cd9c7761dcf28a4cbf3","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-07-21 15:14:23.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"25a773c2bc46982d1ed5e171a40b8462fe26537d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-24 12:29:10.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"5ebf4129eb33dfa5441ddad52fd4e32ef158a579","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-24 12:31:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"66678b2d013cfa68c70b2ebf2312687cde7ab5e2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-24 12:36:38.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"96781ae6a6252cacb25e4b49f06a2806780a0b01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 12:54:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca2694b172fb47a2aacc91d65e64d48d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/010461d3174e4cbda4ff9a8be14808c4 : SUCCESS in 22m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/20a3d8d0f5514a5f8969ad1aec84b057 : SUCCESS in 22m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/528901c7908d480a94b077178a689f82 : SUCCESS in 19m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"93d9f8c2d249bb47b103ff820696748d49182e3d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-24 13:22:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/caaced6fe8314e389650517552f5d054 : SUCCESS in 49m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c9138323f6244a39a9e0d75f98163613 : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":7},{"id":"ad5d476f2bf70eaa825ba3a57c982b270617239e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-24 13:24:36.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950929 : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":7},{"id":"beff0e0d103cf6559f759235f5b808becbf76202","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-24 13:33:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1312/.","accounts_in_message":[],"_revision_number":7},{"id":"f0ce7a5b936c068f1353f7aa1bb84d82547c7d30","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-24 13:36:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1446/.","accounts_in_message":[],"_revision_number":7},{"id":"cc7929e19f695793cd9e40599354cec76589c11a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-24 13:45:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3fbcc419f9e14ef3afa4d17e8b22fafb\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a74934cdec147a782f79741d600ce33 : NODE_FAILURE Node request 200-0000076357 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/88618ec8c2874c18bf51bfb7d263e000 : NODE_FAILURE Node request 200-0000076358 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b7debffd643149b58ffb759548ca667c : SUCCESS in 1h 02m 09s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/23af4db53a9147ca8553bf9952d36a32 : SUCCESS in 1h 15m 55s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dbe4d58e21fd46ebbb9cc13c6f161992 : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":7},{"id":"c9e2ea2b6e102f841e4ec595b992178d8f9262b2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-24 13:53:52.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950929 : FAILURE in 1h 23m 17s","accounts_in_message":[],"_revision_number":7},{"id":"b9d13c1bbe448cf4447316a4949295f3ec5995e8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-24 14:56:47.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-7/FC/ : FAILURE in 2h 25m 40s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-7/iSCSI/ : FAILURE in 2h 27m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"ef7caf01037f0347e3bf51e6a8692565b44e3651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 15:01:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e065edf27364785a78afae0f7b6c924\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfab61ce219046adac66e36bbcdfd6cc : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b6bef73c0ab4f15943dfa77e0ce2cfc : SUCCESS in 10m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f322f59304d4562b21bafad8c95c9ce : SUCCESS in 11m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5170612812564fea835cdfb4bf55f314 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caee0148097e4e4f8b0fb712aed533ba : SUCCESS in 12m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f02d2c47bbba46c4bae19706356b80fd : SUCCESS in 54m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6184b45be43540f9b3e5e5f5d858fe10 : SUCCESS in 47m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6eaaad17d784cde84ef9ccf82d47c6d : SUCCESS in 1h 38m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04c55a43c5da4516a7ee0ae48cd49b52 : SUCCESS in 58m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abb3702660004d638ac83bdc934ac411 : SUCCESS in 16m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/faaa315143c64d32bf68140294758064 : SUCCESS in 22m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/23a134d65908481cbf88378ae54f5f55 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea43ece5608d4a0192c24d55a83b7e30 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d6aa847b0285490db2bfb0e1238e9fad : SUCCESS in 6m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fccd5f9f43da43389b0dd4ab66a882ff : SUCCESS in 8m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/05353e458ae545de972c96f935fc0f7d : FAILURE in 32m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ebe5fccba0144a79a8341f30b1e8e940 : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5d083bc380744304a1fff7ff4d56b531 : SUCCESS in 1h 49m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/57afaf7a290b4622bab91b81ecbc10b0 : FAILURE in 2h 27m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/661eecc603c8470e97b7490ee8128356 : SUCCESS in 1h 50m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e68804e05f844cbd89596af0cb20f822 : FAILURE in 12m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fb4d7fda6e2d48f2bea29a0c70793b72 : SUCCESS in 28m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d45af1153133452ca0f87ec6032e93f8 : SUCCESS in 1h 06m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/469f2d9469bd4ce0880521f4c74acb97 : SUCCESS in 1h 21m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c405240217024d508f331348596086ee : SUCCESS in 29m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a34025c833b94a9a930c715380e9d965 : SUCCESS in 1h 04m 54s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e30ff57599124266a963a53a1bb91628 : FAILURE in 16m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d7b16a7e179c426c879a267b4c837e62 : SUCCESS in 1h 51m 28s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d347cce6540d46b7af2f695de2bd4760 : SUCCESS in 1h 38m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d7ee6affafa47a0bcf23aa0aa3632dd : SUCCESS in 47m 45s","accounts_in_message":[],"_revision_number":7},{"id":"5e0996549692c6aaa1bb5989cd21f6142ef64d57","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-24 15:02:14.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs-rest cinder-tempest-plugin-ontap-nfs-rest : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-iscsi-rest cinder-tempest-plugin-ontap-iscsi-rest : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"0d01ef36ef8eaf30b356bb563d628ca85737ee5f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-24 15:07:17.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_07_2025_15_01_41\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"b2d3081bd48ff074f2598a0c3df1c2cb0b32664f","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-07-24 18:18:03.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ce44cb32b89a62964f2a8f5afd475cfe577abcb6","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2025-07-25 05:49:04.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"25ca334b89db65985c959274e99e274d8b23505f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-07-28 15:11:55.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3f143d84f261f6208d46714c053c768e7f2beb47","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2025-07-29 03:43:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7236f872fbaef4ad3e6b4a451beedbd2eaf745b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-30 06:58:44.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"6efbbd95cd212a77d13fc165e5773f813416e93a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-30 06:58:56.000000000","message":"Patch Set 8:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/61/","accounts_in_message":[],"_revision_number":8},{"id":"f1ffbbbbde4a84c4a92830b2e64f5b35fc0c9e54","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-30 07:08:05.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3a825c211b3dc535c9fb7080737821b78dceba99","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-30 07:08:53.000000000","message":"Patch Set 8:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950929/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950929/8/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"5f512a11088a5b26ca1057456b996688e2725886","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-30 07:09:16.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2228fb536bac4a97d08a675701b2c9f05f11bff8","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-30 07:09:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6aec411664f764462209ee932c1c51ac890e3c07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 07:24:02.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7632437c8594d0880009b21a5aa3287\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f206f2dce57411e87f66faa925920fe : SUCCESS in 18m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f285e292313403f979c0254b977367e : SUCCESS in 16m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b0df2a7fe79a407dbc7868ec7f2527a1 : SUCCESS in 23m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"00f44f6179e6f435cd56b672791373818afce897","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-30 07:45:06.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/60ad67ee42de4522896c508c478b1251\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6b1c9c796148469091cf0986b10ddc58 : NODE_FAILURE Node request 200-0000076809 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2d2e24ee474f45fd9eaccfc74efb9add : NODE_FAILURE Node request 200-0000076810 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/52f7194c117043dabf51a9c784b81d6e : SUCCESS in 44m 31s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a2f43fc4d57d4d04a05f879aed74da91 : NODE_FAILURE Node request 200-0000076819 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1d1208be241647fdb08fc54b0b0b0c11 : SUCCESS in 45m 49s","accounts_in_message":[],"_revision_number":8},{"id":"80a684ea0a07ed281e3c3637fea90426ced5ac82","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-30 08:03:35.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1467/.","accounts_in_message":[],"_revision_number":8},{"id":"575259a91c7576d58a54057d5eec3851d5a20e24","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-30 08:35:40.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f1888c09c4ca46d3874bffa2316e260e : SUCCESS in 44m 59s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/af97dccfd3e2446a829d39e4d2eedcf3 : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":8},{"id":"3cae3a21d8e0e7a71ecb9b98b0654650cc3e5e7a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-30 08:42:18.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1332/.","accounts_in_message":[],"_revision_number":8},{"id":"92886d391cde8904cd7309cccbddaf51590b9162","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-30 08:45:58.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950929 : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":8},{"id":"271136504d5b0211ff7d2f21d32962218749ed47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 08:48:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edaa7c36f70f408ab2070619cdc141b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b7856ea9b10476fb9d7bc0b5c0ad248 : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/278611734e8143bcb6eaa5ccdb944a39 : SUCCESS in 11m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6f613274d2d47bb86a70f988f5b95e3 : SUCCESS in 11m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6cebf6e7cca8425d9c0a538754874694 : SUCCESS in 14m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bae7f71b4404e6c8a1021e3c620adee : SUCCESS in 13m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/aee22c0de46749e8a6b323704fc1b541 : SUCCESS in 58m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2cbfccfa7c6b463797f64ec1adbd57e3 : SUCCESS in 59m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9be6ca34393f421782255fe307b31edb : SUCCESS in 1h 32m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fba0f4e4de54675a18ddc74033ae74c : SUCCESS in 32m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c37579a767bc415baed6b12af1fd7c91 : SUCCESS in 14m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/358df9cda22c4646b80e8eb1aed2b64b : SUCCESS in 20m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1584e6d88a984f458da75001c16aff25 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/54875de51b20405fb56712503b2847c1 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dce1857380ae4a338e8ecdaa5a3d423b : SUCCESS in 5m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/34d6d3b515344d07ad7c76cbbb454fde : SUCCESS in 8m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a53e46d03a904befbd4d72c7680e3bc5 : FAILURE in 17m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7008a1b6b45d48cdae2c078415f8ca95 : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/de5da7036a7e48faab1a75ce5c1f993d : SUCCESS in 1h 09m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d7d5a70a28ba497697bae6b890511d3d : FAILURE in 1h 33m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f9bf949fd81d4586aa792f91d57aa51d : SUCCESS in 1h 45m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9891d1258cb94cde8e60bd2870fe17fc : FAILURE in 18m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/89278388b66842cdb2f6ad219f59ffac : SUCCESS in 25m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/34bb27dc24914a4d9c6ed29c1f067e10 : SUCCESS in 36m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/66faca3c80f64997b7996cf7e4d09880 : SUCCESS in 1h 33m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/39406e677a0245e9b173afea07669444 : SUCCESS in 24m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bebb02433d0d4703917d8f935e2386f0 : SUCCESS in 48m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/96cf3ec2fa354e0fa41aa0b07055f7d5 : FAILURE in 17m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c68abd984d85484891d3ec9c003f616f : SUCCESS in 49m 58s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df3f1b6ae45f42e299952fe7212b50e8 : SUCCESS in 1h 34m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22d820ab815345629ea4c997f7b1e2fb : SUCCESS in 30m 15s","accounts_in_message":[],"_revision_number":8},{"id":"bfb48ffc5643f45f284946481d1dc9e711e848d1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-30 08:49:39.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs-rest cinder-tempest-plugin-ontap-nfs-rest : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-iscsi-rest cinder-tempest-plugin-ontap-iscsi-rest : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"5381ecfd907eee661d60b6cb6d3a02300736d0aa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-30 08:54:19.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_07_2025_08_48_44\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"cf596d5f1271665494524c89c29727174e85cca9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-30 08:54:38.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"9f18d815ce7b7ae560eb307b6f9b25d4b14594f6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-30 09:37:17.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950929 : FAILURE in 1h 24m 15s","accounts_in_message":[],"_revision_number":8},{"id":"c09700515ca8f00ab38bd81a3a34c9366a57e969","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-30 09:51:19.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-8/FC/ : UNSTABLE in 1m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-8/iSCSI/ : FAILURE in 2h 52m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"f5a14e9ca7965790320d483aff0309768ddd3054","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-07-30 10:25:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4d8ea88d89eba82cbb39b56ae55054c0317602c7","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-07-30 12:54:56.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"73f53c62f455314ed3dc023aa93827af58bc8d85","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-07-30 16:47:09.000000000","message":"Patch Set 8: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a95d7b1a73e61da9102955da8a35f431b24454af","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-08-03 05:06:54.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/8/20250803075536/NEC-V-ISCSI : SUCCESS in 5h 39m 18s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/8/20250803075536/NEC-V-FC : SUCCESS in 6h 11m 15s","accounts_in_message":[],"_revision_number":8},{"id":"afd021fa9a134409d6b6dcecce4e566a2a01709c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-08-03 18:50:29.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/8/20250803175955/NEC-ISCSI : SUCCESS in 5h 54m 13s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/29/950929/8/20250803175955/NEC-FC : FAILURE in 9h 50m 31s","accounts_in_message":[],"_revision_number":8},{"id":"942536ae67dde1e8ac3210faf1995ccfdc79f5df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-08-05 16:44:28.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"6e975c88ee852ea8530204e41598b2237585b48e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-05 16:44:36.000000000","message":"Patch Set 9:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/98/","accounts_in_message":[],"_revision_number":9},{"id":"4fbabdb1275b380e1a57335887363945b468ce67","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-08-05 16:46:36.000000000","message":"Patch Set 9:\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"71b37b83bbbad7e68526d14db573c49c9cce22ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-05 16:55:26.000000000","message":"Patch Set 9:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950929/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950929/9/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"a8b3dcb6ddf7cc660ae99d68fa29182c6ea5d1ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 17:14:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4af2a6f037d417f8025f8dd539ad9d0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ffeb3832134b496ba14789d16e3bb24b : SUCCESS in 24m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/684138314c984bfb9e3108caccfc983f : FAILURE in 25m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a85c78ab395547ee9e2182065304209e : SUCCESS in 28m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6012b407358939f84740d5aa3894cf3201e856d1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-05 17:42:18.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b6b9cd3d1e314118b16a28fbc901ecf3\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/608cb7ad801e4ae68d5e3516813004e6 : NODE_FAILURE Node request 200-0000077462 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0e6cd579919c4ef6b2a6af6d40406c2f : NODE_FAILURE Node request 200-0000077463 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/71585b9cd1e04ed2a8bd492672c54a22 : FAILURE in 57m 17s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1c92d20b71b64fa8b82da96dc2d77241 : FAILURE in 57m 17s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c757622e5db94d2081d0da634c379ab6 : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":9},{"id":"0c5744778effb3927fefce88eaf16398a43e9d2b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-05 18:00:26.000000000","message":"Patch Set 9: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c03761672cffd8402900eee265ab6f35e09046bd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-05 18:02:46.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1501/.","accounts_in_message":[],"_revision_number":9},{"id":"2d1ecd2c737d2188ac5fbbeaab0ccadcaa3f54be","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-05 18:20:35.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/de742b7664b243d29f8ec1f064ca4bb7 : SUCCESS in 54m 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2b86c0208ee44178b692ac391cc2c0a4 : SUCCESS in 57m 10s","accounts_in_message":[],"_revision_number":9},{"id":"6efc2ec1dd602c4328bd64609a245ada42a3c5f5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-05 18:32:36.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"00a0303b055b2e1f0f5583485506f4fe06dae646","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-05 18:49:19.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950929 : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":9},{"id":"9858526698f74ec86051b4a7d174f6a688d8ed8f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2025-08-05 18:58:33.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/29/950929/9/ : POST_FAILURE in 1h 14m 02s","accounts_in_message":[],"_revision_number":9},{"id":"b10ec90cdfd1fa90e0a745766b2ab90d0b2bf42f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 19:15:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5ac1d66f9204205a960e7cb01c396a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e87faa8013468bb299f9402b7d0776 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f0cd8a3fc3e4eeda236f3143a9dccf3 : SUCCESS in 7m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f14a00328eda4fe8b28139e32a2b5fa7 : SUCCESS in 6m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b9715c57d5543369423fb9c768395dc : TIMED_OUT in 40m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e862df44740c4fcbb68b8f624b3ef969 : SUCCESS in 7m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/1cb1882239414c81a8b9871b8df8799a : SUCCESS in 28m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/544ef15c65fa4e28ae4d23e148bb14c9 : SUCCESS in 31m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f91b6779a93b430688c3e2e946678046 : SUCCESS in 1h 19m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de9dac7ed40846a5a314f2c309fa0a43 : SUCCESS in 31m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39582be168d9477ca0e8baa0a997f7c4 : SUCCESS in 17m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d01e358a47de4751a384f1a873c1a2e3 : SUCCESS in 26m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cd4ade81d1eb42fcbcb5dd39be09a1c0 : SUCCESS in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/28e54b0d05b342a88e3df3d6628001b2 : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1e8c22ebea7249fabc9ef62fa9b16c2c : SUCCESS in 7m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e2190264928f4983b89ddc1f799ad165 : SUCCESS in 4m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8d9481a7db464691b957ede19a5043d9 : FAILURE in 29m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9854897e380949ca8c98456220b311ec : SUCCESS in 7m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2da860de80cf4854aa3349ec4c291e8c : SUCCESS in 1h 49m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/144a81c0bb244c7fa9db9cae636f741f : SUCCESS in 2h 26m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/efcf6a1efe8c480d9260e704fe0ad251 : SUCCESS in 1h 47m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c94cc61a490a4de9ba5667d6f946744f : FAILURE in 12m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/329414b577fb47489925daa6a64d0080 : SUCCESS in 33m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/45c767ad5b414d7384135e3bed175d58 : SUCCESS in 42m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/93d2335562c24e48ac6f9566c3e97612 : SUCCESS in 1h 44m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3d3b03fdf1c14e708a020f6f2693f0ab : SUCCESS in 45m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cd6cc6aa963e4ceb86f099464d070a14 : POST_FAILURE in 1h 57m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e0a3b52517bd4e11a72d3cef9a8bcb68 : FAILURE in 15m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da7cf8059f5946e8b890ed1ad8beafbc : SUCCESS in 1h 24m 09s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dc664a78e9946af8490e02451df40e4 : SUCCESS in 1h 30m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63f82985869f46b7850ed4cd1cd0fecd : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":9},{"id":"6749a4ed16fb27aa03d861ec28fce798f8222933","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-05 19:20:51.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-9/FC/ : FAILURE in 2h 36m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-9/iSCSI/ : FAILURE in 1h 53m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"10da29a8903812330c4f2bb5a6a9b965c4c20fdd","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-05 19:50:39.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"59ca447955dab2758ff401398c8a5d8049db3399","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-05 19:53:21.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"704b0ba55beaf12c7fe7660d1ed231ec08daad3e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-05 20:23:56.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950929 : FAILURE in 1h 26m 53s","accounts_in_message":[],"_revision_number":9},{"id":"0272e72099c4bf1441cd71fa315e6b992013a62c","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-05 21:09:02.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"93da8b28c00601de9928354eda84a2213b7a7710","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-08-07 05:53:03.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"4ef19ca3552c99bb67206c1f2a12487ce30b1abc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-07 05:53:12.000000000","message":"Patch Set 10:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/116/","accounts_in_message":[],"_revision_number":10},{"id":"ed88fdfb2a6ebe9aab30f924d6b9d1bcf7473f31","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-08-07 05:53:31.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"724b683051701506c8816bb31de973a99ceda204","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-07 05:54:05.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950929/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/950929/10/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"ec3d0756602dccd08b19529719997b76d08bb44a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 06:22:50.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/778d4ef9cfa54908a7c1ac0f8d66f703\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/23f938b538fc456fb7ed12c38df4151c : SUCCESS in 21m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/284e7be5fbd845f292f3cf829d693857 : SUCCESS in 21m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0015f0d25b684b9dbade06ea72638c21 : SUCCESS in 27m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"00a8886c047189339fd728e4fdd77aaaa62537ca","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-07 07:22:45.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4288825712c04202bb36a5f11827f91b\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9a12d5f98af44d359d213a542d15429d : NODE_FAILURE Node request 200-0000077999 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c980708146c49a5b9e8c30b746ac967 : NODE_FAILURE Node request 200-0000078000 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/94ef3f9e27854c57b667a3116d0a1a03 : RETRY_LIMIT in 24s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/739bad3e90d943e3b7f72f9ed59fa620 : SUCCESS in 54m 36s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ed550407b8fa49dc81997e0b1458bcb0 : RETRY_LIMIT in 24s","accounts_in_message":[],"_revision_number":10},{"id":"f0e9ec445087a75cbdb2b22c0979af5e368e8aa8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-07 07:55:22.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"4c34426910c8a4ccbc728342eeed9768b0624419","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-07 08:09:02.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0a40a857e8904e949bd1dab24e6594c4 : FAILURE in 54m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cfc4294aa0d14430a680a6290b86d4c1 : SUCCESS in 48m 26s","accounts_in_message":[],"_revision_number":10},{"id":"4e792fdfd2f62e157f7278e991c9e0ad42e16736","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-07 08:27:34.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-10/FC/ : FAILURE in 2h 34m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-29-950929-10/iSCSI/ : FAILURE in 1h 38m 53s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"dbb72837cb319e3dfa54fc72eac5f40477ee1c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 08:51:05.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d48aee1783e74de99960520f4e8559ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39d01f0500984c958370aa35e1e4b122 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc0e44e8a98e4616a296b0e2637caea0 : SUCCESS in 12m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1085727714144acb88a794eac0adfcd : SUCCESS in 6m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0cf7d7099ce403880d688b67cba68df : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709d2e7bc59c4c17bc8c621ec3dd8d3d : SUCCESS in 6m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/2c2a40e92b8f46048fe2f4e190c4af13 : SUCCESS in 46m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1329645866f1486c84362e6e84234215 : SUCCESS in 47m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2dfe802b6feb41779676fa95249518ad : SUCCESS in 1h 21m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3b472dab13741a684a3d3491a785d76 : SUCCESS in 1h 08m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc85d5e70fa246178604fd5bf3cd94c9 : SUCCESS in 14m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/64bbd5182c5b465c84cba1d29497db14 : TIMED_OUT in 40m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c83fd76840cf4cdcb22ec3514b121252 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8c4712a30de4464690c88304eabf36c8 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c7b21cef553f4bc9a7941951fde6130d : SUCCESS in 6m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a5f35ca64a60454a95df53d691be87c3 : SUCCESS in 7m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/336782ede29b47dea205ab8bec56d9d9 : FAILURE in 25m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b5e71cad5714627862c46732b303ee8 : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1dd1fc15cb6e4bacaddb0de68da76fca : SUCCESS in 1h 44m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f9b6e8a8c64644b093cb64cab0b02305 : SUCCESS in 2h 52m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4a6acc1b455d4466959d1e972623a624 : SUCCESS in 1h 56m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7cd1f920330f4b10956fd2eb3d115a01 : FAILURE in 18m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ee5ba347e9844e4d94656e238eedbd91 : SUCCESS in 33m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2a48e9ffae464c27819607856bdbec3d : SUCCESS in 56m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/72ad488e82f04e128949ba708c8258da : SUCCESS in 1h 34m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5e5bfc5195da4f4a9ba6307e909bd40b : SUCCESS in 38m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5e583c5f22f649299907af46be019ecd : POST_FAILURE in 1h 48m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7e238e40f9b94b58a2d0f88fe3875575 : FAILURE in 12m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e866c8f334254a45a2a4301800f388a9 : SUCCESS in 1h 47m 53s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/217f2b276257473abb24be30a86be138 : SUCCESS in 1h 14m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/874d156fa3e34cb3ab83ed5ab3eb826e : SUCCESS in 29m 50s","accounts_in_message":[],"_revision_number":10},{"id":"8997745c9534890992fe630d690e513318392a4d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-07 08:58:27.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/950929 : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":10},{"id":"7183dcfe51ed06d749998f85d4f90232b6b35663","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-07 09:38:43.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1384/.","accounts_in_message":[],"_revision_number":10},{"id":"75aa5bc259fc9be1927fcf143207f1004dfc368a","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-08-07 09:45:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"88124d22aeb69aa1ea7f4fee080f8850ecbae8b1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-07 10:22:29.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1519/.","accounts_in_message":[],"_revision_number":10},{"id":"7a19d6fcd46feaa20a23a593660cb7a164ae733d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-07 10:38:30.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/950929 : FAILURE in 1h 24m 10s","accounts_in_message":[],"_revision_number":10},{"id":"78176ddeb6ad61225e9b9be9657704957bd3c423","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:00:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2565495c2ed496fb5e861c41f093df3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/089a67f1736c46b6960c6016c73e2b77 : FAILURE in 28m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d7e823a14b64eaa92760f5444b08523 : FAILURE in 29m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4a1e211040894ae7acd065a85bdbb5a7 : SUCCESS in 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a4fe511a409f166e604e183008b799b92a81737b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 12:32:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70e100cf94f84be18a979ef92186616a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7898a86185fe4757a3e8fd184cb2724b : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/963333ad4e4543adbeef0ff1c0279260 : SUCCESS in 12m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/890f11219c82418f976e7c3b216b89fa : SUCCESS in 11m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c457120446cc42618ae0b2414351482d : SUCCESS in 16m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4341070ae21407388eb36c9df236211 : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/6b6f5bc8cebc4b6dba4bd7b2304f0149 : SUCCESS in 1h 04m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b16888d085ed4fd981b2e552246c0574 : SUCCESS in 36m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/10a2e7c1085a48ca92328940d93f9ca1 : SUCCESS in 1h 30m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a9b117a68654283a25d5e554b9b079c : SUCCESS in 1h 08m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/175f8527b5b64426b69eae1ce1267c9a : SUCCESS in 15m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/362f3a51cd474900ad92d831f47fa03d : FAILURE in 25m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/91432a796aa54c02810dbe89260b77c2 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3f29ccb9aa204a2383a9d22c421b0e19 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c11facd1ed94ad2b7ed9ba6395bef83 : SUCCESS in 5m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4d51c9c056fc4e59961c03e55154a266 : SUCCESS in 8m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5e83bd8b1d6d44858f7dbce7d8794e56 : FAILURE in 36m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b5d8781b8d14a89b2cc324e8d71bf51 : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/03db225a43d3451ea298e58ad66f1f22 : SUCCESS in 2h 10m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2b08610717ca4457a0e96668082734e8 : FAILURE in 2h 40m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/355d448bbb6a491a935733d0acdb69f9 : SUCCESS in 1h 44m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3b1fa484e96d456ca7c65f683a6802cf : FAILURE in 17m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/978c59b1bd88467fbd6bf229a7a1471c : SUCCESS in 35m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/264a1da815ad436287d7d15827ecd865 : SUCCESS in 1h 13m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0529f06156474e429f00f318bbabb023 : SUCCESS in 1h 31m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/01af383f9cd84188b0abe6675ed98c8f : SUCCESS in 53m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/574a31dc13724f138223f48b99983f5e : SUCCESS in 1h 36m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/56abcbff1d344a4ab86f27b47141eff2 : FAILURE in 19m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b01a9a92644481e8c93b5b2f2cb8ecd : SUCCESS in 1h 51m 22s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8a0a2036761443e9cce09bb019c7e74 : SUCCESS in 1h 31m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/546cc4d026274aad8aae89821fff5079 : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":10},{"id":"2aef5a79a7c9acd968b3f94b0a5e2c16ab924dbb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-07 12:38:08.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_08_2025_12_32_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"62bcac9f0d2aa9ad748c3946d7bfe4a8cba42652","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-07 12:48:58.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"ccb29f77513537e37c185f2311a0310ad0987987","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-07 14:03:49.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cb7d8b5a2f6e4b109c0bc76f720c5ce9 : FAILURE in 53m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/06d6d3862a0947ecb70fa9361a6195e3 : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":10},{"id":"cf920c621782649d38c3b95c529516cdf5b460ac","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-07 14:39:49.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"1460e35e9e672ca802922c4104d6e58f8ef07d13","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-08-07 19:18:07.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"9c4aaf654fbc7db3454721553e8851734f91b940","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-07 19:18:24.000000000","message":"Patch Set 10:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"8d82766e974c7c105def791da787441e16223ec9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 19:18:32.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"43d93107bfbae788daaf5cf3179e6f553bb7a38e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 21:13:55.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40f942034eb5477c8a606d7281cdd7c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94fa5568e297446bbb6f8e51d7f3c1c0 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7eca4400fab0491a8db7fe8dac06dbad : SUCCESS in 11m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8478cc875cb0448f897773ded33dbf61 : SUCCESS in 12m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2442050c6424f7f9db479bbec9c980f : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/ed9fa36688204c37bf196c7bd1883c42 : SUCCESS in 57m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/52168c49bfcb4f598397870026848208 : SUCCESS in 1h 01m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da69078166fe45139fd20936f097a9d6 : SUCCESS in 1h 23m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a8b2fe8e82941bdbd3d4fb1a0e424e2 : SUCCESS in 51m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1492bc16c9c431db76d9c04ca8b8104 : SUCCESS in 17m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e4adb288556a48c89a1ee8ab7ac7e3e9 : SUCCESS in 1h 12m 52s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/94fba7fe87484fcda24891a37e499a7e : SUCCESS in 1h 49m 30s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ee6eb476f46400bb2e188934f44fdb1 : FAILURE in 1h 27m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e41492294e3c4751b41eb137f0ba3c5e : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":10},{"id":"a4c260d4af9b1f4d8a5ec3268d82b79a34b18ee5","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-08-08 06:20:24.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fcd1eb6ff4470a6c7a8fecee7e2477a5245f09dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 07:13:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d131a208aee4e3baad25301850011d7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/684780ffa56f47c3a9c28bebaa4a047a : SUCCESS in 23m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/418d0ba79a2449bc8573d2e3cef1ca9e : SUCCESS in 23m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/871cf3589645474c84e6183765c79e4c : SUCCESS in 21m 26s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"184392ad321934289876dab0a747c91fac7cd859","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-08 07:14:11.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8f7b6a24940742ea885d060b108d3fdb : SUCCESS in 52m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9450cd8aa8fd49468a13745c911747aa : SUCCESS in 51m 46s","accounts_in_message":[],"_revision_number":10},{"id":"a958cce72f653028a8d6eacdf0705e84b2828f7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 09:03:25.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/987db2937f3a4d08947c3ffaa6352f87\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef90438e067448719ae3753c2ce928df : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2f5b6dd796c4c288d145fe31d5ec951 : SUCCESS in 10m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f24d0c7a5a247389580eb36acc8418a : SUCCESS in 10m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d69f68909834a6bb2d2323d15ceedb2 : SUCCESS in 14m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3c988c0a3b416d860dda09b18552ce : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/87267b8167804be398a43237e04af6ec : SUCCESS in 1h 07m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3c6c6e468df04331bdb3e5b6b54539ee : FAILURE in 1h 09m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd2d80c2833246f49f0dcb4204ec3f03 : SUCCESS in 1h 39m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04107b992ced4e1382d8f84e7222c65a : SUCCESS in 1h 02m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd06eaa97f5743008bed0bfacda87d53 : SUCCESS in 17m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f5d5c7d9306437c9e61fec5df7517e0 : FAILURE in 25m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/463fe6b0416541e08ab47ab96c41eb7a : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ae2a5c64f6994dff9452778460b57b37 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/11f0a3b18efb49a6aeb94e96053ca5bd : SUCCESS in 6m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/eac756089ee44d05ab59a880c43c82e7 : SUCCESS in 9m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/01513d05d7c649e6aafd1485e45f16c0 : FAILURE in 37m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc0ae2c2a84e48b6a2e2d6b60e45ea2b : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f8017c3d108a4b94a2f2ccd9d686a4b7 : SUCCESS in 1h 49m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aa78648374094e1da687d4c63efcccf2 : FAILURE in 2h 36m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/645848be502d4650aa9f882682ab2a8c : SUCCESS in 2h 25m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4924bf98cf214820a7c8a1e347e92b0d : FAILURE in 22m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0b248515ee084efa99cf8c160a65cf1e : SUCCESS in 41m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/962daa8469de4a50a94297bd0e30f5ad : SUCCESS in 1h 15m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1d79b8ed82a14123a3e715c9ccebebd3 : SUCCESS in 1h 30m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c8305c4acc96436cb7a35faeafad3dec : SUCCESS in 45m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a2a8e5d6d8b6452fb018f7473083724c : SUCCESS in 1h 39m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/95a4b655e72c47d9be9f7a05f433c606 : FAILURE in 20m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4fc798d3c929420592682da6f74ed5da : SUCCESS in 1h 44m 54s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cbcb261fda444c0a45f53e0757cd06a : SUCCESS in 1h 44m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c432e10244c743d195a595fa1d6286c6 : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":10},{"id":"510f4c4aa7fce44ad9b40f6e28b5e6426c793661","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-08 09:48:44.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ef52bd73f2904fcaad40185b7f83fe67\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bd6f0395f32e4c63a5ddce8174845835 : NODE_FAILURE Node request 200-0000078319 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b664382acce64b7f843216e6ff9e5f8c : NODE_FAILURE Node request 200-0000078320 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6bcaa14d2a01473d81f6b37b13b443b7 : SUCCESS in 51m 06s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/00535c8c4dfa4f7cb7f4231412063fab : SUCCESS in 59m 23s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8bf71cfd0a8340dc981d52e0eaf70e81 : SUCCESS in 51m 39s","accounts_in_message":[],"_revision_number":10},{"id":"e294011405df953f75658f7f5ef7b3f0e9491152","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-08-08 10:35:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bfa2485d6723db4873e8ee90b6b0cd1584a79b91","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-08 11:30:43.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6e5ce49b90644d418d531695bcc897a2\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/adba5aef83d249b098839e2add5efd19 : NODE_FAILURE Node request 200-0000078368 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b83a9b2f82a48b090b8f5369c48a48e : NODE_FAILURE Node request 200-0000078369 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1e3b466d00404605b1de0c5727ba7b48 : SUCCESS in 51m 27s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9049470c6b2a44d3a06b385a03a9991e : SUCCESS in 54m 43s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b88ef3500d2447b8cb0a1e6031230e9 : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":10},{"id":"5230db3fbea890fae33c559f85c94eb4e601fb35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 11:42:27.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd6bc48dcfff448dad0d1890333124db\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/33e29ca3c73d42a0868dba0f33a10c62 : SUCCESS in 29m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/caff89e73cc1493194bbcebae6d29087 : FAILURE in 28m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4de555e4223947deb1c8eb2fe4dc7694 : FAILURE in 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"29470dfecf24ba3e679d9d3072a77c92664713da","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-08 11:54:11.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c8d0868e7a254badb67ee78cbab5131a : FAILURE in 54m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8a0bdc38aec340e7ba343f8e1c1995cd : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":10},{"id":"13b6be879de94922dbe7a0998acb5484d7e66b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 14:00:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a8750a51670410e96ee209566730625\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a721d9f92ca44c78b312399d862b80f : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8260cd5775524f31b144525169f32dba : SUCCESS in 10m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9324c9da46f845bbae2eef1b8b86b6d7 : SUCCESS in 10m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad113e48602242ce9541251d6f8d1819 : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e70fbb8eaf94b9fa096b5c78583992c : SUCCESS in 6m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/acc48e3d7a1849549119bd4a3be2c0ad : SUCCESS in 31m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/97b4e42b50954e2398e496434b3f63b8 : SUCCESS in 34m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4706f528cb2a4e5bac3f69368dfa8a36 : SUCCESS in 1h 14m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f00769c2dde44ff49f8098f57cc4f2ac : SUCCESS in 59m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b814574efd934abb879514b56276daf9 : SUCCESS in 9m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/950f1938d26143b08ca625365848d0b0 : SUCCESS in 15m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/462b448673914a76b54ef4926010979e : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2fb1013bc33047ee871808f29f9ed8ae : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ba9227be9414d5bb04790c91f2a7b4d : SUCCESS in 6m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e7c106ea3e5744debdf257c65a785536 : SUCCESS in 7m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4cd696d118364a2f88f515a20f1f46e0 : FAILURE in 17m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9990ecf459ac4312a4050c83619ecba9 : SUCCESS in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2cd092f1b1f44d53b8e67e02337db9c7 : SUCCESS in 1h 51m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/495dfd75da8e43519fe3a71ba2727202 : FAILURE in 2h 59m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/40bb49dae95842639fb037d1f3264163 : SUCCESS in 1h 59m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/720d7e7ecdcc4855b7d0a2ddd6741ba4 : FAILURE in 14m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6a52779cf7fb40189f4bc0e828d2d8dc : SUCCESS in 29m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2152e1725b4a43088f8918cb4bfbcf32 : SUCCESS in 1h 07m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c0881ec97935485786d5b03b5e7c6689 : SUCCESS in 1h 36m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/81d047c631fd4786a0895ef61740d8a2 : SUCCESS in 40m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9b336873e48340d3b2d7e2fa08360650 : SUCCESS in 1h 25m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1b4e65c8900046738acc4da49ce0b1cc : FAILURE in 14m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e0e55b4a4be243cfad67d504c3d3fccc : SUCCESS in 2h 06m 43s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d18f536df1b7462483504e623b8754ac : SUCCESS in 1h 39m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e619c9aba95e4697b1a3123146cc4f17 : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":10},{"id":"0d69adf2a4a64ccb77bc0855a768c228f4ddcb61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 14:00:57.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"fed552b59bf710060bfb5ba5f47c0d42c8ee2c75","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-08 14:05:45.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/08_08_2025_14_00_38\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"3e145cb8d57d30d626961390c6d4d22fa02daa30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 16:03:04.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ef8318240fe4b59bd47636dcbae6248\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efc006c9c5414adabdb4d22b3ab24b90 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cc2527e44c045758ba807b20ee6de1f : SUCCESS in 7m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45f3ae1991c740808d1a020c5e62c16e : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44503a32d46c4f85b4cba510288247ad : SUCCESS in 6m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/1c7bac3f3f384640967318a504fab256 : SUCCESS in 34m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/22939c2820074e83b72575a21a3b6eaf : SUCCESS in 1h 14m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d682da1b7d047b1ad92b6339de0f92e : SUCCESS in 1h 20m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/327717db3527488f88ade7065ec16320 : SUCCESS in 31m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55181bb884864dc39d8c0ad80a43d935 : SUCCESS in 16m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/742a165585d24005896161085ded1eb3 : SUCCESS in 39m 15s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b6ac8b1f55684b77a6da266f1edef1d8 : SUCCESS in 1h 44m 29s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28a76e5f5af34eb485cf27b6cca43cf8 : SUCCESS in 1h 39m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1c5cb509c82408387493c6cd982fc19 : SUCCESS in 28m 59s","accounts_in_message":[],"_revision_number":10},{"id":"57267f9c3ad3ff0be70dc6a4b4636af2cdc62a00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 16:03:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"3ad90c9a13ec87116e1acead31da2fadf663a9ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 16:04:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2faee7fbbad46178a79bb4571b8080f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e8bfa1b8a264e12822505b3a5ade341 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b9a970545ad49ac9e54f7590d37af4f : SUCCESS in 45s","accounts_in_message":[],"_revision_number":10},{"id":"3d8fc3a6b1927f1e89cb8dae4176240dc17a91ce","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-08 22:04:12.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/207ef61519f94697973701039fdb2c0f\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/7423ff77534b4a56a49044f821823f35 : FAILURE in 11h 18m 39s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"fb8349c2e0c4b9d8610651318ccfd53e07ff84e1","revisions":{"1cc93619c1d017ea7eec90073aa8bf37d11f2fa5":{"kind":"REWORK","_number":1,"created":"2025-05-27 05:19:27.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/29/950929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/950929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/950929/1"}}},"commit":{"parents":[{"commit":"18d93ae3dabeb8c92fa3896a91593a407ce50baa","subject":"Merge \"Replace deprecated datetime.utcfromtimestamp()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18d93ae3dabeb8c92fa3896a91593a407ce50baa"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-05-27 04:55:17.000000000","tz":-240},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-05-27 05:18:00.000000000","tz":-240},"subject":"NetApp - Fixed detach issue for multi-attached volume","message":"NetApp - Fixed detach issue for multi-attached volume\n\nVolume with multi-attached type can be attached to multiple\ninstances. Added the logic to handle the removing of cinder\nvolume from multiple instances.\n\nClose-Bug: #2110274\n\nChange-Id: Ibb4d71868106226b5513e4c825f769008a446727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1cc93619c1d017ea7eec90073aa8bf37d11f2fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1cc93619c1d017ea7eec90073aa8bf37d11f2fa5"}]},"branch":"refs/heads/master"},"6909d2ba524e0bb3ae1d9d723cdf8dbce206d5f9":{"kind":"REWORK","_number":2,"created":"2025-06-09 15:48:16.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/29/950929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/950929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/950929/2"}}},"commit":{"parents":[{"commit":"deef0d4d28d6a1ba84b9f3af49efdd85042f68ff","subject":"Merge \"[SVf] : Enable support for mirror-pool option for replication volume-type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/deef0d4d28d6a1ba84b9f3af49efdd85042f68ff"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-05-27 04:55:17.000000000","tz":-240},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-06-09 15:45:55.000000000","tz":-240},"subject":"NetApp - Fixed detach issue for multi-attached volume","message":"NetApp - Fixed detach issue for multi-attached volume\n\nVolume with multi-attached type can be attached to multiple\ninstances. Added the logic to handle the removing of cinder\nvolume from multiple instances.\n\nCloses-Bug: #2110274\n\nChange-Id: Ibb4d71868106226b5513e4c825f769008a446727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6909d2ba524e0bb3ae1d9d723cdf8dbce206d5f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6909d2ba524e0bb3ae1d9d723cdf8dbce206d5f9"}]},"branch":"refs/heads/master"},"c0d8daa8398e0037bc18425b6af5219cec895f11":{"kind":"REWORK","_number":3,"created":"2025-06-09 17:13:21.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/29/950929/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/950929/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/950929/3"}}},"commit":{"parents":[{"commit":"deef0d4d28d6a1ba84b9f3af49efdd85042f68ff","subject":"Merge \"[SVf] : Enable support for mirror-pool option for replication volume-type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/deef0d4d28d6a1ba84b9f3af49efdd85042f68ff"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-05-27 04:55:17.000000000","tz":-240},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-06-09 17:13:03.000000000","tz":-240},"subject":"NetApp - Fixed detach issue for multi-attached volume","message":"NetApp - Fixed detach issue for multi-attached volume\n\nVolume with multi-attached type can be attached to multiple\ninstances. Added the logic to handle the removing of cinder\nvolume from multiple instances.\n\nCloses-Bug: #2110274\n\nChange-Id: Ibb4d71868106226b5513e4c825f769008a446727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0d8daa8398e0037bc18425b6af5219cec895f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0d8daa8398e0037bc18425b6af5219cec895f11"}]},"branch":"refs/heads/master"},"a2e9d41d0bb072057c408ac634fe6b86834fd31c":{"kind":"REWORK","_number":4,"created":"2025-07-08 11:11:36.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/29/950929/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/950929/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/950929/4"}}},"commit":{"parents":[{"commit":"3380f01ff2876a6fdb6d4fe94984936277aefe90","subject":"Merge \"Hitachi: Fix to have compatibility of GAD volumes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3380f01ff2876a6fdb6d4fe94984936277aefe90"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-05-27 04:55:17.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-08 11:11:23.000000000","tz":-240},"subject":"NetApp - Fixed detach issue for multi-attached volume","message":"NetApp - Fixed detach issue for multi-attached volume\n\nVolume with multi-attached type can be attached to multiple\ninstances. Added the logic for FCP/NVMe protocols to handle\nthe removing of cinder volume from multiple instances.\n\nCloses-Bug: #2110274\n\nChange-Id: Ibb4d71868106226b5513e4c825f769008a446727\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2e9d41d0bb072057c408ac634fe6b86834fd31c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2e9d41d0bb072057c408ac634fe6b86834fd31c"}]},"branch":"refs/heads/master"},"bf4c65b03d4b752fa0acb4e8635144e036d49d69":{"kind":"REWORK","_number":5,"created":"2025-07-10 14:59:48.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/29/950929/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/950929/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/950929/5"}}},"commit":{"parents":[{"commit":"21af57108f9f67ce9e5af78a6b7eef0d3f828576","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/21af57108f9f67ce9e5af78a6b7eef0d3f828576"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-05-27 04:55:17.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-10 14:59:24.000000000","tz":-240},"subject":"NetApp - Fixed detach issue for multi-attached volume","message":"NetApp - Fixed detach issue for multi-attached volume\n\nVolume with multi-attached type can be attached to multiple\ninstances. Added the logic for FCP/NVMe protocols to handle\nthe removing of cinder volume from multiple instances.\n\nCloses-Bug: #2110274\n\nChange-Id: Ibb4d71868106226b5513e4c825f769008a446727\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf4c65b03d4b752fa0acb4e8635144e036d49d69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf4c65b03d4b752fa0acb4e8635144e036d49d69"}]},"branch":"refs/heads/master"},"e27058def7537efe79b568e1f11f73e1eaceaf1f":{"kind":"REWORK","_number":6,"created":"2025-07-10 15:20:36.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/29/950929/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/950929/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/950929/6"}}},"commit":{"parents":[{"commit":"21af57108f9f67ce9e5af78a6b7eef0d3f828576","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/21af57108f9f67ce9e5af78a6b7eef0d3f828576"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-05-27 04:55:17.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-10 15:20:23.000000000","tz":-240},"subject":"NetApp - Fixed detach issue for multi-attached volume","message":"NetApp - Fixed detach issue for multi-attached volume\n\nVolume with multi-attached type can be attached to multiple\ninstances. Added the logic for FCP/NVMe protocols to handle\nthe removing of cinder volume from multiple instances.\n\nCloses-Bug: #2110274\n\nChange-Id: Ibb4d71868106226b5513e4c825f769008a446727\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e27058def7537efe79b568e1f11f73e1eaceaf1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e27058def7537efe79b568e1f11f73e1eaceaf1f"}]},"branch":"refs/heads/master"},"0da02a810ea8151a85c0223e642ef018308f1480":{"kind":"REWORK","_number":7,"created":"2025-07-24 12:29:10.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/29/950929/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/950929/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/950929/7"}}},"commit":{"parents":[{"commit":"21af57108f9f67ce9e5af78a6b7eef0d3f828576","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/21af57108f9f67ce9e5af78a6b7eef0d3f828576"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-05-27 04:55:17.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-24 12:28:55.000000000","tz":-240},"subject":"NetApp - Fixed detach issue for multi-attached volume","message":"NetApp - Fixed detach issue for multi-attached volume\n\nVolume with multi-attached type can be attached to multiple\ninstances. Added the logic for FCP/NVMe protocols to handle\nthe removing of cinder volume from multiple instances.\n\nCloses-Bug: #2110274\n\nChange-Id: Ibb4d71868106226b5513e4c825f769008a446727\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0da02a810ea8151a85c0223e642ef018308f1480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0da02a810ea8151a85c0223e642ef018308f1480"}]},"branch":"refs/heads/master"},"0215931370daf70651760398a9524ae6f2be7859":{"kind":"REWORK","_number":8,"created":"2025-07-30 06:58:44.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/29/950929/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/950929/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/950929/8"}}},"commit":{"parents":[{"commit":"21af57108f9f67ce9e5af78a6b7eef0d3f828576","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/21af57108f9f67ce9e5af78a6b7eef0d3f828576"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-05-27 04:55:17.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-07-30 06:58:27.000000000","tz":-240},"subject":"NetApp - Fixed detach issue for multi-attached volume","message":"NetApp - Fixed detach issue for multi-attached volume\n\nVolume with multi-attached type can be attached to multiple\ninstances. Added the logic for FCP/NVMe protocols to handle\nthe removing of cinder volume from multiple instances.\n\nCloses-Bug: #2110274\n\nChange-Id: Ibb4d71868106226b5513e4c825f769008a446727\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0215931370daf70651760398a9524ae6f2be7859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0215931370daf70651760398a9524ae6f2be7859"}]},"branch":"refs/heads/master"},"f1f69cffe0834880d3f49b69c5dabe1307d9f43c":{"kind":"REWORK","_number":9,"created":"2025-08-05 16:44:28.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/29/950929/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/950929/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/950929/9"}}},"commit":{"parents":[{"commit":"256e26e54d71caf16f1aecc4df6b670d814ba18b","subject":"Merge \"RBD: Fix issue with managing volume with type properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/256e26e54d71caf16f1aecc4df6b670d814ba18b"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-05-27 04:55:17.000000000","tz":-240},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-08-05 16:44:11.000000000","tz":-240},"subject":"NetApp - Fixed detach issue for multi-attached volume","message":"NetApp - Fixed detach issue for multi-attached volume\n\nVolume with multi-attached type can be attached to multiple\ninstances. Added the logic for FCP/NVMe protocols to handle\nthe removing of cinder volume from multiple instances.\n\nCloses-Bug: #2110274\n\nChange-Id: Ibb4d71868106226b5513e4c825f769008a446727\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1f69cffe0834880d3f49b69c5dabe1307d9f43c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1f69cffe0834880d3f49b69c5dabe1307d9f43c"}]},"branch":"refs/heads/master"},"fb8349c2e0c4b9d8610651318ccfd53e07ff84e1":{"kind":"REWORK","_number":10,"created":"2025-08-07 05:53:03.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/29/950929/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/29/950929/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/29/950929/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/29/950929/10"}}},"commit":{"parents":[{"commit":"256e26e54d71caf16f1aecc4df6b670d814ba18b","subject":"Merge \"RBD: Fix issue with managing volume with type properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/256e26e54d71caf16f1aecc4df6b670d814ba18b"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-05-27 04:55:17.000000000","tz":-240},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-08-07 05:52:15.000000000","tz":-240},"subject":"NetApp - Fixed detach issue for multi-attached volume","message":"NetApp - Fixed detach issue for multi-attached volume\n\nVolume with multi-attached type can be attached to multiple\ninstances. Added the logic for FCP/NVMe protocols to handle\nthe removing of cinder volume from multiple instances.\n\nCloses-Bug: #2110274\n\nChange-Id: Ibb4d71868106226b5513e4c825f769008a446727\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb8349c2e0c4b9d8610651318ccfd53e07ff84e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb8349c2e0c4b9d8610651318ccfd53e07ff84e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
