)]}'
{"id":"openstack%2Fcinder~952731","triplet_id":"openstack%2Fcinder~master~I4a0bb1b9fd7df101e04ad13266bc5c4d4331d09c","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"36737":{"account":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"last_update":"2025-10-17 17:37:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a0bb1b9fd7df101e04ad13266bc5c4d4331d09c","subject":"[doc] Update Volume List API doc for metadata param","status":"MERGED","created":"2025-06-17 01:50:12.000000000","updated":"2025-10-17 17:38:17.000000000","submitted":"2025-10-17 17:37:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952731","meta_rev_id":"6ad5618dba2f3175d68375865b66b4b61cd439cc","_number":952731,"virtual_id_number":952731,"owner":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-17 17:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-10-17 17:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-10-17 17:29:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-17 02:31:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-17 02:50:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-06-17 03:24:24.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-06-17 03:39:11.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-19 14:11:32.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-06-20 20:37:57.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-30 03:13:21.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-06-30 06:10:11.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-10-17 17:29:03.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"b1f2993c47436cc1e0742757c5a20404cff88366","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2025-06-17 01:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f91cbcfaa65c9b1fc6039545212ddbd72f943292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 02:31:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/524c50c1931048c397a9f533b7f47c8f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/88fa5361b9dd4eec8b63734699130ad8 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ca9baa79834d0c8d7ab5021513f87f : FAILURE in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe4093e4f714167845bcf781f50dbc7 : SUCCESS in 12m 41s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/348acfab55e9477698f3473e48779986 : SUCCESS in 3m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3324c4ff9a33480182a8ec516b2b772e : SUCCESS in 5m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c867450ce3a242e9b17307c695f314f8 : SUCCESS in 9m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ba3dcc1d389a4c9fbcdf8c60912e670c : FAILURE in 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17b8a2f5f622a443c12c135c50a27107063d2e4a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-17 02:50:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/988/.","accounts_in_message":[],"_revision_number":1},{"id":"30aa702f3c03f8ca406fc8a3d27eb6cba6eb441c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-17 02:50:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1118/.","accounts_in_message":[],"_revision_number":1},{"id":"ca733b5725e8706988e213c369124fcf253f2f9b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-17 03:24:24.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-952731-1/FC/ : UNSTABLE in 2m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-952731-1/iSCSI/ : FAILURE in 1h 33m 48s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"750009e7bf979e2f75425599237d7d3df8004034","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-17 03:39:11.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"580f2e7710753567837d8efc5d0ce66d8dacbab8","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2025-06-17 09:37:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c79401cf643aafd0eb7b4b21d8c9ee2027ed6645","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 10:07:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57e9e701c811407aa5959f10409dc214\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/42680673db5a4a9bb3f79ba9b0cb7607 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/700f8a780b9c4f77b721433738b21095 : FAILURE in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e126d4a09260471c8343f4fd8d7f8b6f : SUCCESS in 11m 07s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2e2ee0a28ec1484da4f1e8203c5168c4 : SUCCESS in 5m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9add16fc6cdc4cc5a124a3fe5c3c7a7c : SUCCESS in 4m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9294e3d0ef674faea3281659967174b6 : SUCCESS in 7m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d05b99f271734a50a48e541692897f34 : FAILURE in 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6cbb057fe350b7f61597b822036781fc72052b3c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-19 14:11:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/952731/1/20250619175932/NEC-V-FC : SUCCESS in 4h 29m 32s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/952731/1/20250619175932/NEC-V-ISCSI : SUCCESS in 5h 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d922c97809c696689f766136e0df7861c53e9b15","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-20 20:37:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/952731/1/20250620232127/NEC-ISCSI : SUCCESS in 5h 52m 33s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/952731/1/20250620232127/NEC-FC : SUCCESS in 6h 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a6e6f7254c874ce822250587df9640ae3edfc7f8","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2025-06-25 23:46:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"656c42d29ebc743360092e5fba31bf9d3a4561f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 00:26:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c9d989a6dfb495c9637058cb72d35a7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c91ff932c7ab413c9ccfbf6b78dfd3c2 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de51ff450c92411fb76de186d8bc58aa : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ca2a7fa4524e3b9a5f8bd1aabeaf1c : SUCCESS in 13m 55s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/79711e5165624ae38fec9a80005422a1 : SUCCESS in 4m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/898e113ddf4743429758958f71bd617d : SUCCESS in 8m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c8f7d781334242e6a1ff105bd616822c : SUCCESS in 7m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/edac35d40da44391815e2b0b460e4e21 : FAILURE in 38m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a139f8b334192ae22bdd0d49cdfb67ec0376c8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2025-06-30 02:19:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"71c1a10a630eaa4143f26294b26a79803da4b37b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 03:08:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2021e12a7ed04d63a7a4342549ea14b1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5136641c84df4bd48acff96f11f85734 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/398a3d0ab35d4e6e9d7c66f39a4a6416 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c91bf7970c74b8582ee685d7571b313 : SUCCESS in 15m 56s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/08279576cbf24a5c8700296dcf992205 : SUCCESS in 5m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a983067f85f1419b80e7fa8bbf4146cf : SUCCESS in 6m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cb43c72daa53442f8ce568c93b7fca27 : SUCCESS in 7m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2b3f88e4c3ab4cfcb4667e4ca21ba36e : FAILURE in 34m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87a9cd69de3b2ccc2f1d2550f8f7b13521f205fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-06-30 03:13:21.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_06_2025_03_08_43\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"3c6f2d5cac0cfdd23b77e9e69e7f793cac76105a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-30 03:20:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1233/.","accounts_in_message":[],"_revision_number":2},{"id":"8fdd356291d5f6c8e10a9de486b6ca50948dd322","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-30 03:53:03.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1104/.","accounts_in_message":[],"_revision_number":2},{"id":"011d3e2aeda8873d0ab981fde204555c1c978874","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-30 04:06:17.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"3c9c6baca40d8ac3e782243c54797855e8812d05","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-06-30 06:10:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"a93b2d13a2fc1bccde05d7558cc751120a0bb32e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-04 13:05:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/952731/2/20250704154908/NEC-ISCSI : SUCCESS in 5h 50m 37s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/952731/2/20250704154908/NEC-FC : SUCCESS in 6h 15m 52s","accounts_in_message":[],"_revision_number":2},{"id":"eb164ffd16145a9ef88b9859fb043da7182ddfd5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-11 06:56:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/952731/2/20250711102624/NEC-V-ISCSI : SUCCESS in 5h 25m 28s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/952731/2/20250711102624/NEC-V-FC : SUCCESS in 5h 30m 15s","accounts_in_message":[],"_revision_number":2},{"id":"be54d2b602de5ff1cfc3261a08ff989880c48c3a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-17 17:29:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7de4c8911fa456ad2f5cf7659fb07f9c2cadebe7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 17:29:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9959dbc06ef01a17fc61ee5e6bcfa7f38ff0624b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 17:37:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05bd2174bb89437dbae6b7794c4477a5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/873f28f2877b426896a67c5d90e4d703 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70b8814875fa4ea18477ba144b47452d : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e887f10367446d8b25ba9361ec7b2b1 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f8bd135abb8a76cfcb879863a2d9a93a81abd0d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 17:37:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6ad5618dba2f3175d68375865b66b4b61cd439cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 17:38:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f1336a14f504517aad7ab6a2b2378ba\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/dacbfacc356941dabab05951fc5c70de : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dc904f2423a49e6a3511e287397ab67 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5889a1c3c4fc179b4365a059399ae7e0acb8904","revisions":{"0aa00d1283e2b48ecc88554cd5f48048cd37c433":{"kind":"REWORK","_number":1,"created":"2025-06-17 01:50:12.000000000","uploader":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"ref":"refs/changes/31/952731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/952731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/952731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/952731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/952731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/952731/1"}}},"commit":{"parents":[{"commit":"111e43b9c05c3a7ac8e14771d976bc6c68a70d9c","subject":"Merge \"tgt target: Provide unique scsi_sn and scsi_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/111e43b9c05c3a7ac8e14771d976bc6c68a70d9c"}]}],"author":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2025-06-17 01:41:43.000000000","tz":540},"committer":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2025-06-17 01:41:43.000000000","tz":540},"subject":"[doc] Update Volume List API doc for metadata param","message":"[doc] Update Volume List API doc for metadata param\n\nAdd a metadata query parameter to the \u0027List accessible volumes\u0027 and \u0027List accessible volumes with details\u0027 APIs.\n\nCloses-Bug: #2114777\nChange-Id: I4a0bb1b9fd7df101e04ad13266bc5c4d4331d09c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0aa00d1283e2b48ecc88554cd5f48048cd37c433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0aa00d1283e2b48ecc88554cd5f48048cd37c433"}]},"branch":"refs/heads/master"},"d5889a1c3c4fc179b4365a059399ae7e0acb8904":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-30 02:19:50.000000000","uploader":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"ref":"refs/changes/31/952731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/952731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/952731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/952731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/952731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/952731/2"}}},"commit":{"parents":[{"commit":"5c6faf9c67e5c8bae03f085eadf8f87d45e5b4eb","subject":"Merge \"NetApp - Fixed Device busy error when multiple cinder volume clone created in parallel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c6faf9c67e5c8bae03f085eadf8f87d45e5b4eb"}]}],"author":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2025-06-17 01:41:43.000000000","tz":540},"committer":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2025-06-30 00:09:39.000000000","tz":540},"subject":"[doc] Update Volume List API doc for metadata param","message":"[doc] Update Volume List API doc for metadata param\n\nAdd a metadata query parameter to the \u0027List accessible volumes\u0027 and \u0027List accessible volumes with details\u0027 APIs.\n\nCloses-Bug: #2114777\nChange-Id: I4a0bb1b9fd7df101e04ad13266bc5c4d4331d09c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5889a1c3c4fc179b4365a059399ae7e0acb8904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5889a1c3c4fc179b4365a059399ae7e0acb8904"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
