)]}'
{"id":"openstack%2Fcinder~952904","triplet_id":"openstack%2Fcinder~master~Ia143e729a3e84e22fcecf6f5bac8b0b3ae311553","project":"openstack/cinder","branch":"master","topic":"bug/2114993","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-07-16 22:19:47.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2025-07-16 22:19:47.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2025-07-07 16:59:16.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-06-26 08:56:24.000000000","reason":"\u003cGERRIT_ACCOUNT_36180\u003e replied on the change","reason_account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-07-16 22:19:47.000000000","reason":"Change was submitted"},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2025-06-26 17:22:25.000000000","reason":"\u003cGERRIT_ACCOUNT_30555\u003e replied on the change","reason_account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2025-06-26 14:24:40.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-07-16 22:19:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia143e729a3e84e22fcecf6f5bac8b0b3ae311553","subject":"NetApp: iSCSI/FC detach operation fails when multiple initiators connected","status":"MERGED","created":"2025-06-19 15:10:56.000000000","updated":"2025-07-17 21:28:28.000000000","submitted":"2025-07-16 22:19:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"952904-bug/2114993","meta_rev_id":"3df1bffd4d43af64555ee991b406d472172eb4b4","_number":952904,"virtual_id_number":952904,"owner":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-16 22:19:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"recommended":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"all":[{"value":1,"date":"2025-06-26 08:56:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":2,"date":"2025-07-07 16:59:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2025-06-26 14:24:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-16 19:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-16 19:45:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-19 15:11:54.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-19 16:05:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-19 16:18:38.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-06-19 16:40:15.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-19 16:55:08.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-06-19 17:00:47.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-06-19 17:02:10.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-19 18:19:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-22 01:52:58.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-06-23 15:02:28.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-06-25 04:05:52.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-06-25 11:33:11.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-06-25 11:33:11.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-06-25 11:33:11.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-06-25 22:34:03.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2025-06-26 11:37:26.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-07-02 12:54:41.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-16 19:45:16.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2025-07-16 19:45:44.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"5983e438a04ac2985392f61475971bffeab0b34a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-19 15:10:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b25a1d3364ae8b347a55142132a4775e614f891","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-19 15:11:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/69fadeb4ad704db2ae7c8508bc0b4ca0 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fad5254091b1432db13009f087c0dcdc : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"d0e7c1e4b2f1a4802ebfba4518a9fcc120de6268","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-19 15:12:18.000000000","message":"Topic set to bug/2114993","accounts_in_message":[],"_revision_number":1},{"id":"5561dbbc08cafcb3fe5f122308c188aceb6d4c0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-19 15:15:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9c767f375fb4377f7b9ca016c133297b300d0a41","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-19 15:16:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fa45664a85ec44cd9a89b2842c9640b5 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/aa18b57868de4cab8117e7d9ba98390d : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"c45770d3c64d9b9164b32e7afd2d6ac88249fa6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 16:05:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79b76b348b0748b584bf8ee8b5f91113\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5b6026ecc8ad4f749e675af6ef23578a : SUCCESS in 27m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/73ce4d3e34f44bb8903cc6842050afd8 : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/16b5fab3e1764c19a22a1df54e4d001f : SUCCESS in 31m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6bcd281bee99018a94c7cb77af403ca364bb40c7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-19 16:18:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f545c088aaea4c2ab3ab8f5e8b31be49\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0726125e1be54577a87c2c4d2ba2cbdf : NODE_FAILURE Node request 200-0000068470 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/039954787cd748f183326bbc1dbc73ac : NODE_FAILURE Node request 200-0000068471 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/469e79de07cb47df92b65e9a046b155c : SUCCESS in 1h 02m 58s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/42ec341dd1f14820aea7eeba0b65a536 : SUCCESS in 53m 03s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c2aba41be0ed494994aa626ba1ff0288 : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":2},{"id":"c2713bb45b90a7a39767626befc5b37ae30ad524","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-06-19 16:40:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/952904 : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":2},{"id":"abed464625d22322f75aa27d726f04bb23486466","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-19 16:55:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1007/.","accounts_in_message":[],"_revision_number":1},{"id":"947c34693558e2754b87365e980c2d860d2621a9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-19 17:00:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1136/.","accounts_in_message":[],"_revision_number":1},{"id":"5ed2156b218c716579db6fa4524e14e8c7385024","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-19 17:00:47.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-952904-1/FC/ : UNSTABLE in 2m 4s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-952904-1/iSCSI/ : FAILURE in 1h 49m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"4aca665bd75d2d50e39e3a4c9a76d0e4fd14a9ff","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-19 17:02:10.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b457c3bb8c9dcb6f44402d633cfd3d5a17b1e8a6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-06-19 17:08:58.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-952904-2/FC/ : UNSTABLE in 2m 32s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-04-952904-2/iSCSI/ : FAILURE in 1h 53m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"eb2bf743d52b950248657b2a44b92b98cfdaf25f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-19 17:52:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1008/.","accounts_in_message":[],"_revision_number":2},{"id":"dfdf741167d59a299085dda9a20bf42c69ba4582","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-19 18:01:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1137/.","accounts_in_message":[],"_revision_number":2},{"id":"acef0f26a3538127242e44a6306c0ec668c8eb2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 18:19:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0904e981a4384d85bf03bb0478177d13\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/204129b17f194c328d094d26187fdf10 : SUCCESS in 6m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfa5060948874156b260cfde0fee04f9 : SUCCESS in 16m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2d5208e4e43495995e2ed8ca3bfb4fc : SUCCESS in 9m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd1d2fbd1b6c4f19a3523781fd4a0302 : SUCCESS in 16m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbb24e577154d2ebf232ea9e50cf3dd : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/5203fbb8daf7479b8a1ce42871f939c1 : SUCCESS in 48m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6bba6d5f1bf443edb5767e75a31fd950 : TIMED_OUT in 3h 03m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e1b73c8f30a048cb95fccb23c2c91426 : SUCCESS in 1h 19m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ba0dce4d09945859b0cc8147df60b74 : SUCCESS in 1h 04m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a3fe4b91fa184ed1a8cc36b275f82271 : FAILURE in 26m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/74a744822cf54edea82f86d1c831abb9 : SUCCESS in 5m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0560c473ece048b79c6ce387e9637444 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0fb6c55f6dc64268aabb9306f765bfd2 : SUCCESS in 5m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a4dd87fbc80045cba3de3a5b5ba0fe11 : SUCCESS in 8m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eb0eea5a38bc4970ae159885e4eb6510 : FAILURE in 39m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a5732d88d4d4c8eaf2ec002e86c458f : SUCCESS in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc9ada0510b742c59b1736bccfd600e4 : SUCCESS in 1h 47m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c8574899482d4a4e8fcea0ad64cf83c3 : FAILURE in 2h 02m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/60d36302599b46b7b15d3eaadadc273e : SUCCESS in 1h 37m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/47c8cf1183504e309363da4607613790 : FAILURE in 20m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6511594b27a4417db01daa3a2152331e : SUCCESS in 28m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e178bb901e82402cbb28951e91e7b056 : SUCCESS in 1h 04m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0bc8dcdf7f4c4f6c872863dc90dcb634 : SUCCESS in 1h 23m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/817cad4ebfb74412af49f5f439813ec3 : SUCCESS in 46m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2a32f9a40cb54f39920e2946fd519125 : SUCCESS in 1h 22m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3e3cbf8b15a1486a82de95e0f7563717 : FAILURE in 16m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c18f33485504cbcaf8bc286a727567c : SUCCESS in 1h 35m 51s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12afe33b5127441381b88744371f6cf1 : SUCCESS in 1h 35m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38479e6af75b48c0bbc5b5cb464d3d49 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":2},{"id":"6913b1f07f974d72a4bff5eb923212e8e8a7a0c6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-19 18:40:05.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"93bbf00b3bc6f1e88730ba670206f0e6a4923440","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-22 01:52:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/952904/2/20250622063049/NEC-V-FC : SUCCESS in 4h 06m 18s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/952904/2/20250622063049/NEC-V-ISCSI : SUCCESS in 4h 22m 06s","accounts_in_message":[],"_revision_number":2},{"id":"be81824e36c66424234aa1843e4160b4b8371943","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-23 04:53:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"547294697eed54db2872f1d2c313e83cd665cd1e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-23 04:54:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4869aae524ef45d7be6c6bfd1fe64203 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2c4dd5bd3f864f5a951e349750af8b93 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"f2fd4cf29bb68fa97814dd7a5c7061256bcf3228","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 05:11:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec65aaa1500940da95883a9de37ad4a1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a25d91bac04c4378addf56aaadf4e09e : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f946210a0eb84285b1f5f91a16cf4c34 : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/33e70a48123d4325822d3f72b0af7d6f : SUCCESS in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d01e9134d6e4fc22bdc0100e97526866c3206ce9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-23 05:40:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e2f0de1347814d47960657f1d1fa1a6a\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0398eed7265c4086bf35684e493e2c4a : NODE_FAILURE Node request 200-0000068682 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fadbcaf5ea7145718612e483e95bb509 : NODE_FAILURE Node request 200-0000068683 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/03139af33852472d8a1b5b0b2ab61d11 : SUCCESS in 44m 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ef5dbac92f8443a58955a7f2dd3f3c6f : SUCCESS in 47m 00s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/34396b9779e84c1f90e04dce3e3b8794 : SUCCESS in 45m 27s","accounts_in_message":[],"_revision_number":2},{"id":"b8f516517790dfaedb32590fbfcf46d2c16e0724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 07:05:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5fb3c6c9fe248c8b0b491e441c40b99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0beb9a8d2bc46fdaa007200fad0b956 : FAILURE in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d57dc9b6c9f493abe54d7307401ab08 : SUCCESS in 9m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6265589648b4edc8ff324368601b998 : SUCCESS in 12m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7cdcdc0235648208392b6973b480261 : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/111e7131ce1d4efaac4a4241bd5e5367 : SUCCESS in 16m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/e0249f7cc6ba43eba370e32c893e0d3e : SUCCESS in 59m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b4de47407b4540c88a17bcb725ead40b : SUCCESS in 1h 00m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d841682353d4ec4a682e6a0143c1425 : SUCCESS in 1h 28m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a580c29446d542b9835de0a140f0abea : SUCCESS in 56m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4357e4e1c04f41a38b0b951d9d044f71 : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e4bb31f545aa47c1918cf7ec03a84232 : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d1b64b6514cf4cf997d606e506d3ae94 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9124d87ec8da4aab9877047996c713b1 : SUCCESS in 5m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/15a3f6a1e52c4c0fbc9430bee2dce8eb : SUCCESS in 7m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/19e6053973cf4a16aa7e8a3218bff047 : FAILURE in 26m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f062fd5a74cc4d8389e31ade73844496 : SUCCESS in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a439fb6d362947479c9a11c7ba6b8102 : SUCCESS in 1h 28m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bad7d2928c5744639d996c7a20485070 : FAILURE in 1h 52m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eb7bbd2f95aa4865a1538d344c7fd8d8 : SUCCESS in 1h 24m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e3c1e3b12e454a63b3a1a0a346e9788f : FAILURE in 11m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/744b6edea0af44d89e225994a64dd39e : SUCCESS in 26m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/acf1fd0193284591a1f787e3c0f72cbc : SUCCESS in 57m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5d01334ee1eb4889bdbb5b4023315dca : SUCCESS in 1h 12m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8b634ecb38e84aec95c059f9d25edf29 : SUCCESS in 31m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9b423305a300495cb931cd6b4a312047 : SUCCESS in 1h 15m 27s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bd4b6c322a1b4d9a99b35d29e2d3b921 : FAILURE in 19m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fbb50a1bddd940c08c301bbc7078863c : SUCCESS in 1h 34m 08s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49fc7f6310714c389a2842768501b681 : SUCCESS in 1h 32m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bee09f08f2584b8c8eefe7b12ffdb2d4 : SUCCESS in 57m 09s\n\nWarning:\n  Comments left for invalid file cinder/api/api_utils.py\n  Comments left for invalid file cinder/api/common.py\n  Comments left for invalid file cinder/cmd/backup.py\n  Comments left for invalid file cinder/cmd/volume.py","accounts_in_message":[],"_revision_number":2},{"id":"86a360b2b23f99947164a616340b9cdfca845394","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2025-06-23 15:02:28.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"976bb5e7feddbd7be8fc237d04569ce988faf0b1","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-24 03:38:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a1aa0a18a129abf9d5bc4f88ac77b5d79924169f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-24 03:39:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c22dd321169445d2babafbd79dc6a339 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fc157e105c474120b472ae57817d1853 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"17148dee1de2a88d4ffc61c62e62aa14f5b2fbbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 04:08:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5facd56d607b409cbf47dc14ec35311a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/150bbbc5c4754f82952f46a4cb148acf : SUCCESS in 21m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e496617ef5084430b4858ce019b0524b : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2b03b68cd92642a5be0c5501f9d3b651 : SUCCESS in 27m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"29fe0363183d95e063ce21e6b44ae2cf5fc0a056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 05:30:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15f9830ebccb4ba2ab99f87e7cb4763c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51b808785d8248e1a58a78523b21ec75 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da6b68d42edb41a9a3e031cea8b3cdf8 : SUCCESS in 6m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/089e54d477084321ad0314817c8a9ee9 : SUCCESS in 10m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27117939fc4c48c7b370b57454a65487 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c77eaf651548d9aa25d19976297029 : SUCCESS in 14m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/3b122cd7fbec41029725015f6f8bded5 : SUCCESS in 1h 01m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b8bc6bd394e440949a191b976111b3b8 : SUCCESS in 1h 03m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e538f130375d4e1387bf506934b6f091 : SUCCESS in 1h 19m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0510c7c2ece43e69232bd289db8cba6 : SUCCESS in 1h 04m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8108e73446864f3ebfb28bcbbd8ea322 : SUCCESS in 22m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/70c4aba2f7724a46b77edd653381f9b8 : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/815ea0e59a3748e28509282b097cc24b : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a949a57f1be349c6b0379c8838b975c8 : SUCCESS in 5m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/12a6ce0fa0c34a7092e9dbc1051bdddf : SUCCESS in 8m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/26b671a16f304b5db508913a213494df : FAILURE in 32m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3fe36b53de4549e68ada34497bbba76d : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/37034e5ea5c742b89c8635d620d69166 : SUCCESS in 1h 49m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f961eb61f2914cd58816be63378433f5 : FAILURE in 12m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a770d8f15be14d7cb5f8cd132e5e8575 : SUCCESS in 1h 46m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3e8692557b7c49bab6a7267a9662dcec : FAILURE in 19m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dbe48dc5130c4aa7b6a7f44258b26c7c : SUCCESS in 32m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ec9cda3cf1c04b55af13094d962bb0da : SUCCESS in 52m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8cd89f4c8f584e0c8f53d579be34189c : SUCCESS in 1h 41m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a6cf338118eb49a78c44314d0a38912b : FAILURE in 10m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b330d92d52404fec884544d4ac9d0900 : SUCCESS in 1h 26m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8fd206097cae4120a1672ef171b93b01 : FAILURE in 17m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c33065a66c204219b4c4ac84079be6d7 : SUCCESS in 1h 26m 38s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4c96fd2d2c840258da95758e163b733 : SUCCESS in 1h 16m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dc7bc5ee72e4524ab728b08a9e6703b : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":2},{"id":"83717ee7568a3dbdeb7f423e7b38c6321d097845","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-24 05:44:14.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3b35ae0c09394c7da3b5971f6656ee9b\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9c94dc69018348158a32588e7de69bc9 : NODE_FAILURE Node request 200-0000068744 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ed0087b6d66145efbbe0ec18aba6b975 : NODE_FAILURE Node request 200-0000068745 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3f77cd6ea416476e9256832c0d6e1268 : SUCCESS in 48m 41s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/80c6af0331a74bba966842e45ba2994f : FAILURE in 2h 05m 29s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/188e6ce4e4eb4004931dbf9037c6c561 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b2f0e5b0172c13c38739f06ffe17fede38975dfa","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-06-25 04:05:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e11ed4f4fef5312d73744943ae504c34d79503b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-25 11:25:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3beed20a3f79383a4a814e3efec438de326d28a7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-25 11:26:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1d02f706560d423288ba3419c4b05306 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ceeed6fe8e9a469c961578f1eb59b0b7 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"0fa334ea08db75f9ccfa5eb4c7e1742bbe46c1ed","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-25 11:31:36.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"73334c037ad0d6b637ab6ee86d35ca131e3e96cf","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-25 11:33:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"90f8f29504051cdc5daed140cc621836cbc81f95","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-25 12:23:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cd5bc045b35745f0a3f91f4f31e3449c\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e70d3c4390f04d4c9a589f343ed4dde9 : NODE_FAILURE Node request 200-0000069345 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a0c8354e3ce440699dced9713370f9ea : NODE_FAILURE Node request 200-0000069346 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/630e1ab8b53143aabd377381e65b1e1c : SUCCESS in 47m 04s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca6fd68cefa84e739baba4a4faeba1bf : SUCCESS in 57m 10s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a96d736ddbb0407e99618ea357c16dba : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":3},{"id":"bba37a29c3a0b16f0e2821318bdecea693a47c00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 14:04:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5842069cb3354acabe1269a736bcef81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9aa1e8e9c6d46f297e2c37954e32103 : SUCCESS in 7m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4214b86b48ca446e8e2c85ce1d630c90 : SUCCESS in 10m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a68c421fa7941d783f38f9ded8c6ca8 : SUCCESS in 12m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96fbbeed6e9d4e678c0ed0f361f3ed0e : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/438fbbe25e8342fc9b6aa42ff78e1a8d : SUCCESS in 12m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/504bafab528d4d77b636082a76cdeda1 : FAILURE in 9m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/04cb6eb9cc424a8cbdfd4d4713afc9c9 : SUCCESS in 1h 09m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28b29d6c9273490dae853f8316a7830a : SUCCESS in 1h 40m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a97897f866748c683d44ca24c95f2b9 : SUCCESS in 57m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64da261b37504a7b8ef9445861aaf72a : SUCCESS in 13m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7f9ae8c1f9f4402b883b0d78b1843fa4 : TIMED_OUT in 40m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0995cd4c755e4f21b27b04316e026281 : SUCCESS in 3m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7d72b590426140858532f9b74e974913 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1a41813dd14f4479a001b5a0cdbbe103 : SUCCESS in 5m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/55454d67145846a3859ade6318d32194 : SUCCESS in 10m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cbf7b5b0529448f19ea7141d7cd1a60a : FAILURE in 37m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3b41e219d008495b8db62e4b333b6e22 : SUCCESS in 6m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/528d8f5426bd4eadb1cb56805ab0c5d5 : SUCCESS in 1h 28m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c09e23bda13048448014d5561aa8b872 : SUCCESS in 2h 36m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b0371172ad9646b699fb021f04e4317c : SUCCESS in 1h 51m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c4b84df81ab8487b8d06502214c42ac3 : FAILURE in 18m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e5606dab2dbe4c33b0a88b90f997bfdc : SUCCESS in 37m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7715c8e039a54f51bfccaee1ff141dad : SUCCESS in 1h 07m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/04f7ffec3e7842d2aba3d4c99ad6bb00 : SUCCESS in 1h 53m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d1c07bc82f7d46eeabb5720a7c752e45 : SUCCESS in 38m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e8bd64a2365d4fc3a6e8df4e40868ab6 : SUCCESS in 1h 27m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9e55c207bf6f4185b6637f779577e41d : FAILURE in 14m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3a9e487ec36e4475bdcb3f4e2e362b97 : FAILURE in 11m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72e6f8bffdcf4e9593de12305c8a05ab : SUCCESS in 1h 31m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e9b8994890d43acb6d0e6c6087dec4f : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":3},{"id":"675e53357ac208b4a329cdd18888c0d48c894ea0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 14:25:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a917a4598561402b9c277888314987f3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/141d30c2ad654a94a53aea6c45060acf : SUCCESS in 29m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/87043eafd51f4e0aa92b8ee5ba2e3496 : SUCCESS in 31m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb02f4173cbb4abbb0c1cacb88638987 : FAILURE in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4b4b9b450a8fc11d9994027ee23fc23699e82446","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-25 15:17:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd3aebd5664bbef3784ec3ee9e1553641cd26333","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-25 15:25:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fa6e41959bf14e0c95485697153a3427 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7e8dc64db2b6449db9be47bef6cfbaee : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"bc7cad00a7f870dd2dde8a25a67e16c37d4eca84","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-25 16:18:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6e7bbfc9c0d442bebcc7f51584bcf309\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/954966f43b95437185c1dda1151cb2ea : NODE_FAILURE Node request 200-0000069385 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/83cd4ce1f7284f5abc45fadc53e76e16 : NODE_FAILURE Node request 200-0000069386 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c0dde638bc84e99bfcf7351481a9a93 : SUCCESS in 54m 04s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f41878a164d34e21881b2bffb92e91a4 : SUCCESS in 51m 28s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/40506bfece924fcd86824233cfb16740 : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":3},{"id":"03da09a50b97bb10ffba62eac138750633c7797c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-25 16:22:16.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"a0bd401515ae230aa520929ebae5f11ddb2540a0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-25 17:02:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1194/.","accounts_in_message":[],"_revision_number":3},{"id":"05eeda202a184afcbf87da4137e03aaee254dc0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 18:11:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5d03e095ed646518abb0186351b816c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d90dc0a992e4481e89f68686033a36d0 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55e3570ae6f2424a9b7fe1502ce5f9af : SUCCESS in 10m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2aec0361e64d4bab9bf3e051c23f3aff : SUCCESS in 10m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5d2dff1c5624522bdb3f8eb02e95a5e : SUCCESS in 16m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24826c9bf63f4714a17fd9343fb20156 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/395301963dd04763a484b9a011e6a61d : FAILURE in 9m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e42dbf05955466c89b876ec7d71f4d6 : SUCCESS in 1h 04m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f7102105c0874c7db06c8298cee61bf6 : SUCCESS in 1h 39m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cedd83c07c3747b599106dc66654d056 : SUCCESS in 32m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79e062d9249e4d8ebde8165a8247df8c : SUCCESS in 17m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce2a760559944f8bb29616dc2250e893 : SUCCESS in 20m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/350b2ba6f9d04c04a20e87225abd9230 : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/15923b7f8ca84e4f894328d69aec2376 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/891edc75288845768100ccaaec16e951 : SUCCESS in 5m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/650ee15720f943f6a81ed1427c451c54 : SUCCESS in 8m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/78b9a7344ebb4c608cfdcafe9776b5e0 : FAILURE in 9m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/20ed7f24a122462dbaf03d9335545750 : SUCCESS in 4m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/891a92f71d5a46649ee94be6d31219be : FAILURE in 10m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/040e70be2fbe4597b4f35f69ded941f7 : SUCCESS in 2h 31m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5c7f249851d348aeb9acc44f8c313a2f : SUCCESS in 1h 43m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dc2a441c00ed421f82709d70ceb1c8da : FAILURE in 14m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dc3723dfd377454b9fe380c359b99162 : SUCCESS in 31m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d492fab0b21d49deb6f9ae89dca0ebc6 : FAILURE in 15m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/88a9177a87db4532b064999cf177af9c : FAILURE in 10m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/defb59b9e4224c2ebb460c47cefa4014 : FAILURE in 9m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6a74c85fca5b48dc8100681c6ae60f40 : SUCCESS in 1h 39m 27s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6da6c17fd571478896ff2aa81103acef : FAILURE in 9m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bdbe88fb25b842f8a5b3fe521f035c7d : SUCCESS in 1h 38m 01s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06e443e91f1b4ef7b080e4414b3fb122 : SUCCESS in 1h 45m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65e2a654b0cc4e8c9412c4ae4a6c0c4e : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":3},{"id":"94e91a1a37e30a5d7d40add17a983b8097552290","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-25 18:21:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1065/.","accounts_in_message":[],"_revision_number":3},{"id":"4c9588ab246ac972211ff3ffafddcea07784bd3c","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-25 22:34:03.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c99f6d7b8e201976872b58860ce0795879d210c5","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-25 22:42:15.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bb4765cc959fe59158d26fb380c1eedc3fd43360","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 23:02:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50568d309ea0403d8804c722609df987\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ed5191bf8e34a8c95d66bce51de1c71 : SUCCESS in 23m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/979104b2ad104f99bf575c8282badde9 : FAILURE in 29m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e12f8db664d64469a426f29593e9d9a4 : SUCCESS in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6d0d9dfab6d28bdb4c74a4ba3923743d8bba7800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-26 03:23:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4b0003ecda13f777e3028a0135c5bf70ffb9a25e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-26 03:24:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/70bd0387a07d44fca1ca685b9c19f6a1 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c1d98d7fc15544a4bc492bd1c9233262 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"91de429ecb44ed4fcbfe076be8055742238ca88f","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-26 03:31:40.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fe7fc9e0aff63c09a07efb95c3647b11f7789788","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-26 04:15:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/49f1bfee4e7846359c90e3ead3f8cbd8\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ba9a3ddcaefb4e02a599c24d4b6f91d5 : NODE_FAILURE Node request 200-0000069447 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2bf4b2e8da224967a29490aea5a35215 : NODE_FAILURE Node request 200-0000069448 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c2e12a036174402aa155428a339f5673 : SUCCESS in 50m 45s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d1704a2f64084cb6bf82d302c47efc4f : SUCCESS in 51m 48s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7bc5dd3f87dc40fcb32434d66608bb53 : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":4},{"id":"604bf4dd79d90ee287b08b45963653b01a66c668","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 04:20:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/812a60aa6be9420780c142b450b11052\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8bcf029cb944695a9f3faa4da8dad4a : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2b87f8c7b56a4df78aa45e9833d14d73 : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/70922a2f846246dc86134876991fd94d : SUCCESS in 28m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f00733556ec32283379ba7340b20bd651834e3eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 05:19:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb85d34c5ed94fbbb434920bc7e47d4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52ed8439e31e4c64a690442d5d0b8794 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/314b8858c7d445f6b637aa6ac590d552 : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44c1f3e58a4e4c4c98d8213a88c0184b : SUCCESS in 10m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/602bee1d0b134b4e839d1984f2453c2b : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26764a5afa954a9f86e74d395f62e3b9 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/f2bcae3f96b54d299aeb5a4939b1de99 : SUCCESS in 46m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38c61d41226c41218f440159806d11f5 : SUCCESS in 59m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4aafc28bdf844389b7e5e2f0855315e : SUCCESS in 1h 17m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d22f98048f0e4815ba12fc7cdd84b53c : SUCCESS in 32m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55d672fe4d914330819cc23f73b34301 : SUCCESS in 18m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2a52aa11c3e3454595b09af26226cc5d : SUCCESS in 28m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6dccad9eeccc4f68a13d5698ca425498 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9b8f6a38bf354ca4a409ca78dc8bc173 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d9b6158d31fc477885ab1072af9e9011 : SUCCESS in 5m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dc43653589784823b624d72f6434a1d9 : SUCCESS in 8m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4453e502ff164d1faa80f0a1a6e7c747 : FAILURE in 29m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fb504b0a4c7346ea91b323207c47ea01 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/be3416ee90ab47db85754f8c846b47f3 : SUCCESS in 1h 06m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3f38bfe2c9ff46bc97e9e3514005e38f : FAILURE in 1h 54m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b18b029a557c4cc68ff2ba6717c375fb : SUCCESS in 1h 42m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/369506e7fef64a3cac0eb526d2858701 : FAILURE in 13m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f87c494bb05b4b4381c7dcd08c307dca : SUCCESS in 17m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e19e8841dd9647e4b73ca3ac3278c20e : SUCCESS in 1h 05m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/20f1499c61184884a2515a0ff8a3e351 : SUCCESS in 1h 13m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/682db0d17bae42fd928e15a851c5efdf : SUCCESS in 39m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ba2ee9f6b25645888ede595a3e60474f : SUCCESS in 1h 12m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e722d3a97cb4466f9be45550a43c1f4e : FAILURE in 13m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a4328b9e393497e9b923e9d38b7bdf5 : SUCCESS in 51m 27s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83062283754f4022b1e6e4c53f7819fc : SUCCESS in 1h 33m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7954e2521a24738b1d2837c936d045f : SUCCESS in 43m 51s","accounts_in_message":[],"_revision_number":4},{"id":"e25641cf0a9324dae773b9715612f06442fe7024","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-26 07:56:05.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"befa3cf63a1376e8d2c46871e5a2f5fc4fca26e5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-26 08:46:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1206/.","accounts_in_message":[],"_revision_number":4},{"id":"ebd09cd7e64ba7430f165be6384f374960ca9309","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-06-26 08:48:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"16aebc9fe30f902994b55a4b0456418377831b88","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-06-26 08:56:24.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a16da2070ff78c5e2b2cb29c4010cde2787f4319","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-26 09:58:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1077/.","accounts_in_message":[],"_revision_number":4},{"id":"faacbc0b07925dcf4476e0acb687c2a46d32c1d6","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2025-06-26 14:24:40.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3269de6282c0c710de90c4c44c5f90f13e635a44","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-26 17:22:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"160a05ffa46c03c06a9a9acf51fb361ed43045fd","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-26 17:27:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3b2b4dd2c0a0f62d2d1a61b9fad4fbaa30b86440","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-02 12:54:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/952904/4/20250702153701/NEC-ISCSI : SUCCESS in 5h 47m 51s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/952904/4/20250702153701/NEC-FC : SUCCESS in 6h 17m 36s","accounts_in_message":[],"_revision_number":4},{"id":"e57b6d560a52223aac3ac8649780b9faae9d8a55","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-04 08:34:04.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/952904/4/20250704004546/NEC-V-ISCSI : FAILURE in 16h 14m 04s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/04/952904/4/20250704004546/NEC-V-FC : FAILURE in 16h 48m 15s","accounts_in_message":[],"_revision_number":4},{"id":"185b865413b113192cba3d850c7025f1213de23e","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-07-07 16:59:16.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"91cef6ad4a960a4537fd05be5edcc6a85a03b58d","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-07-16 19:45:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"60dd68aa4f086df30bf46a840904fa1f17972d0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 19:45:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f9bee300b2b83772fb5270254c27cf1c0714b9f2","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-07-16 19:45:44.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"266b7e7d9925d1fdeadb6d2f4972570c1b042d2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 22:19:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27357a2755a947089883603a2e14daf5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d64dca6e38b445b89c524914adab327 : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b65e206bd6354d1ca6e81cda878fe2cd : SUCCESS in 9m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6bb3eb222ad94070ac3e9fd8f16fd53b : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1e77d268cbd4ea494ff9997f1fcfd80 : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/55521405e8ed42b39145bd2e15bb3199 : SUCCESS in 50m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9b10ff2bcd7d47339e9d80a96534c19a : SUCCESS in 1h 14m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55eb7dc3a22f477395fff23234b6e230 : SUCCESS in 1h 30m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c98fae5a88924efbbd5ba33a724906e3 : SUCCESS in 50m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76c534945a3e4d56808ae16730bc555d : SUCCESS in 16m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/760d4033c82c4e2abc047bdbf7445a0f : SUCCESS in 1h 09m 18s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e086599b5943493f9436b3316053894f : SUCCESS in 2h 05m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddb871cba6e54c2b909b3c21cd675fd4 : SUCCESS in 1h 16m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6e3e98f8f674b08ac48ff2b2942b1de : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":4},{"id":"533473ed2f6ed510d0cac9d4a44445d6684125aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 22:19:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6f261df253aa8e9f7aefef1d41c522152d5d7f8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 22:21:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/311257f7801649ca8011a5358c9cfcae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5daed2ff32fe40afb50a5f11a903b3fa : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbc3ed68d6054c5d820cb727e1354963 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4},{"id":"3df1bffd4d43af64555ee991b406d472172eb4b4","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-07-17 21:28:28.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/6c66868b469d40ab999803d793da74a6\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/c05dc42074834213be3c7048fe267133 : FAILURE in 17h 14m 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4addad4baf3d6978d6fc597737bdd7f8fe697479","revisions":{"fe6885d3f33a485ecdb78f7988f0d9d4cdd6c4ff":{"kind":"REWORK","_number":1,"created":"2025-06-19 15:10:56.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/04/952904/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/952904/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/952904/1"}}},"commit":{"parents":[{"commit":"c948b22eace9f6aa299a0af3ac374864ff8ff163","subject":"Pin flake8_import_order version to \u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c948b22eace9f6aa299a0af3ac374864ff8ff163"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-06-19 15:02:15.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-06-19 15:10:52.000000000","tz":-240},"subject":"Netapp: iSCSI/FC detach operation fails when multiple initiators connected","message":"Netapp: iSCSI/FC detach operation fails when multiple initiators connected\n\nDuring the detach operation of iSCSI and Fibre Channel (FC) sessions,\nthe process fails when multiple initiators are connected to the Device\nUnder Test (DUT). Investigation reveals that the root cause is related\nto improper handling of initiator identifiers due to the use of a space\ncharacter as a delimiter.\n\nSpecifically, the system expects initiator IDs to be parsed and processed\nas distinct entities. However, when multiple initiators are listed or\npassed as a single string separated by spaces, the parsing logic incorrectly\nsplits or merges these identifiers. This results in failure to correctly\nidentify and detach each initiator session, causing the overall detach\noperation to fail.\n\nTo fix the above, updated the delimeter from space character to comma.\n\nCloses-Bug: #2114993\nChange-Id: Ia143e729a3e84e22fcecf6f5bac8b0b3ae311553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe6885d3f33a485ecdb78f7988f0d9d4cdd6c4ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe6885d3f33a485ecdb78f7988f0d9d4cdd6c4ff"}]},"branch":"refs/heads/master"},"571f5ed7e36a9730258b0708f17208f0502e3cb4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-19 15:15:09.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/04/952904/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/952904/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/952904/2"}}},"commit":{"parents":[{"commit":"c948b22eace9f6aa299a0af3ac374864ff8ff163","subject":"Pin flake8_import_order version to \u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c948b22eace9f6aa299a0af3ac374864ff8ff163"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-06-19 15:02:15.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-06-19 15:15:09.000000000","tz":0},"subject":"NetApp: iSCSI/FC detach operation fails when multiple initiators connected","message":"NetApp: iSCSI/FC detach operation fails when multiple initiators connected\n\nDuring the detach operation of iSCSI and Fibre Channel (FC) sessions,\nthe process fails when multiple initiators are connected to the Device\nUnder Test (DUT). Investigation reveals that the root cause is related\nto improper handling of initiator identifiers due to the use of a space\ncharacter as a delimiter.\n\nSpecifically, the system expects initiator IDs to be parsed and processed\nas distinct entities. However, when multiple initiators are listed or\npassed as a single string separated by spaces, the parsing logic incorrectly\nsplits or merges these identifiers. This results in failure to correctly\nidentify and detach each initiator session, causing the overall detach\noperation to fail.\n\nTo fix the above, updated the delimeter from space character to comma.\n\nCloses-Bug: #2114993\nChange-Id: Ia143e729a3e84e22fcecf6f5bac8b0b3ae311553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/571f5ed7e36a9730258b0708f17208f0502e3cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/571f5ed7e36a9730258b0708f17208f0502e3cb4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7d8da036d22ba4f53ac82d28de3a87fc647114f2":{"kind":"REWORK","_number":3,"created":"2025-06-25 11:25:41.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/04/952904/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/952904/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/952904/3"}}},"commit":{"parents":[{"commit":"c948b22eace9f6aa299a0af3ac374864ff8ff163","subject":"Pin flake8_import_order version to \u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c948b22eace9f6aa299a0af3ac374864ff8ff163"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-06-19 15:02:15.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-06-25 11:25:18.000000000","tz":-240},"subject":"Netapp: iSCSI/FC detach operation fails when multiple initiators connected","message":"Netapp: iSCSI/FC detach operation fails when multiple initiators connected\n\nDuring the detach operation of iSCSI and Fibre Channel (FC) sessions,\nthe process fails when multiple initiators are connected to the Device\nUnder Test (DUT). Investigation reveals that the root cause is related\nto improper handling of initiator identifiers due to the use of a space\ncharacter as a delimiter.\n\nSpecifically, the system expects initiator IDs to be parsed and processed\nas distinct entities. However, when multiple initiators are listed or\npassed as a single string separated by spaces, the parsing logic incorrectly\nsplits or merges these identifiers. This results in failure to correctly\nidentify and detach each initiator session, causing the overall detach\noperation to fail.\n\nTo fix the above, updated the delimeter from space character to comma.\n\nCloses-Bug: #2114993\nChange-Id: Ia143e729a3e84e22fcecf6f5bac8b0b3ae311553\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d8da036d22ba4f53ac82d28de3a87fc647114f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d8da036d22ba4f53ac82d28de3a87fc647114f2"}]},"branch":"refs/heads/master"},"4addad4baf3d6978d6fc597737bdd7f8fe697479":{"kind":"REWORK","_number":4,"created":"2025-06-26 03:23:05.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/04/952904/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/952904/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/952904/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/952904/4"}}},"commit":{"parents":[{"commit":"c948b22eace9f6aa299a0af3ac374864ff8ff163","subject":"Pin flake8_import_order version to \u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c948b22eace9f6aa299a0af3ac374864ff8ff163"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-06-19 15:02:15.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-06-26 03:20:15.000000000","tz":-240},"subject":"NetApp: iSCSI/FC detach operation fails when multiple initiators connected","message":"NetApp: iSCSI/FC detach operation fails when multiple initiators connected\n\nDuring the detach operation of iSCSI and Fibre Channel (FC) sessions,\nthe process fails when multiple initiators are connected to LUNS.\nInvestigation reveals that the root cause is related to improper\nhandling of initiator identifiers due to the use of a space character\nas a delimiter.\n\nSpecifically, the NetApp\u0027s REST API  expects initiator IDs to be parsed\nand processed as distinct entities. However, when multiple initiators\nare listed or passed as a single string separated by spaces, the parsing\nlogic incorrectly splits or merges these identifiers. This results in\nfailure to correctly identify and detach each initiator session, causing\nthe overall detach operation to fail.\n\nTo fix the above, updated the delimeter from space character to comma.\n\nCloses-Bug: #2114993\nChange-Id: Ia143e729a3e84e22fcecf6f5bac8b0b3ae311553\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4addad4baf3d6978d6fc597737bdd7f8fe697479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4addad4baf3d6978d6fc597737bdd7f8fe697479"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
