)]}'
{"id":"openstack%2Fcinder~953157","triplet_id":"openstack%2Fcinder~master~I43d1846c3cacc6bd84655d56f29077f1a642baac","project":"openstack/cinder","branch":"master","attention_set":{"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2025-06-24 15:11:18.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I43d1846c3cacc6bd84655d56f29077f1a642baac","subject":"Revert \"Pin flake8-import-order\u003c0.19.0\"","status":"NEW","created":"2025-06-23 22:53:33.000000000","updated":"2025-06-27 18:25:39.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c0fba11592a012d2428991f25ae8b3b023007d85","_number":953157,"virtual_id_number":953157,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-24 16:36:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-23 22:56:24.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-23 23:34:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-24 00:49:03.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-06-24 02:03:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-26 09:10:30.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-06-27 18:25:39.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"98b32d3f2336abd968e9b3cf8647d86061cf357c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-23 22:53:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12c298bc413cd84143c29e32b1ea598f9b478828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-23 22:54:12.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"aff7a5aaab48ba2286e5f4a129511d14f46d8d88","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-23 22:56:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b2e63b8f2c5949ff8c7d7807de823cf2 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d05e83eba7c44d5786e2858850732237 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"8d7c3fbb691fda87780e272ca148208ac2c9a668","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-23 22:56:55.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e5bad38fc396f3240c0dabd649bd294043fc5c0c","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-23 22:57:25.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1\n\nCopied votes on follow-up patch sets have been updated:\n* Code-Review-1 has been copied to patch set 3 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\").\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47d086ba88ae157b1ff454b4cc0c843c85e6ce99","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-23 22:58:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e959d205a0d14a8da8379e3a5cbd3b8b : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/952a1c58031144afbed683d71b49a0eb : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"515ac22ce39317070c12f4a667335b075ee6c2c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 23:34:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff87f067e8944fe4985c82ff388863d7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eab38a8403ec46f08e7d3a73628b0025 : FAILURE in 27m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b61ca4358a34c81a6d50dea0a8a8f36 : FAILURE in 27m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0c1c5743309b442d8558e482d8944a3f : SUCCESS in 35m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"66fa16fa8465315b6228d79c64dbb9098392944f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 00:49:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1028/.","accounts_in_message":[],"_revision_number":1},{"id":"57685458d7728f0f8a5e1c540c637a41796c877c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 00:50:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1157/.","accounts_in_message":[],"_revision_number":1},{"id":"fbe08a8509f8eeafd21fbb2546e2f3d6fd042aa7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 01:44:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1158/.","accounts_in_message":[],"_revision_number":2},{"id":"66e61473564a4a1bf2857d16ebe7aea554b8f005","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 01:47:43.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1029/.","accounts_in_message":[],"_revision_number":2},{"id":"40fa4830a5f232ff4daae02d39d943fae29699f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 02:03:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22f7d90efac3493b97886d5a4a5e5ad3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4ccb2797a32a4ffd86f4bc77eea9e36d : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af83d6aadf947c1b62dbb4e4b96ed98 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f48d148402224e0a90d73f78d116ff99 : SUCCESS in 8m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1d4a541a0694cb086fa554211b0894b : SUCCESS in 8m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a480076d6b324092ba0c9cdb0d73c4d4 : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e3fc88869b4912bbcf48aeaf12660b : SUCCESS in 7m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4527c26ad707486680a7aabc6e5a3f41 : SUCCESS in 3m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/36c143cbd39f41f8be953162889d6e4b : TIMED_OUT in 3h 04m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba6204f595c54ff88dabf0f9838719c8 : SUCCESS in 20m 19s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bd11fadf736843fc9775b12b6d5fa1a4 : SUCCESS in 6m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/49f4711473ac4b99ac5a22c49d2f0765 : SUCCESS in 5m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0be576a0c3f24e27a8b2bfbd2a558a48 : SUCCESS in 4m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/09262a05e1794904990279709f5ac5ef : FAILURE in 37m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1324e232127c479294420976a4b2c3ef : SUCCESS in 5m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7feabbf919a84dae9ef8110966fe3429 : SUCCESS in 1h 26m 20s","accounts_in_message":[],"_revision_number":3},{"id":"6c1043f236b94d070f4b73a669ff3322a8b75dea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 04:35:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1161/.","accounts_in_message":[],"_revision_number":3},{"id":"82abd0f1031118733b14e1092f6e8c4ba3d54a13","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 04:37:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1032/.","accounts_in_message":[],"_revision_number":3},{"id":"6621dcac49da166cb2060e5d0e500870c77465d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-24 13:13:54.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c0abad6ac217587114d2635b833c80e08dcb7395","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-24 13:15:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/49c2970b67024af89c8d8dd734708406 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4a57d74059574a3399439af1655fa228 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"0d70e05547dd2b138f2d2aa3e21ca1320efec455","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 13:48:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d82e8bc7c94c93bb0c290ca4ee4c41\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f83ee6bd8cde479c86fef8d23e808acc : SUCCESS in 18m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/869ab34029ed4ad88ad9fd0d546464a2 : FAILURE in 32m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/36d127bead8b449a9b515c9de2901c90 : SUCCESS in 21m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7f6cbb3d6075a60b2090267e5f1bdbde40c751b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-24 14:02:09.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"34afb14ee832eea6a3c7c6fab95c780a033df483","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-24 14:03:21.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/03b4c45a3ef8413e83ec2d0d8153de2a : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9d61f37aba774557901bc0ba7a56044f : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"7a953d06c5856559d3ce1b8d0a4e8d32a6704403","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-24 14:20:41.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"3631a60b96c62fb7e0c1586fb91fdbb8f6e918d0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-24 14:21:42.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0ba941a0cf7a475c8f27e69427178783 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/298191ac1c4447b6be4e6a317a0e3b5a : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"a93ba5e2f03a2903b413631658975b7e10307693","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-24 14:22:45.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"c7c3dbf62a8f3cbe36fbcab6f8983274e253c371","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-24 14:23:52.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d88bbaf58bc0432080ae7ba7a2af9cbc : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c334be2df1e5449fb4dc74e863d9f722 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"dde0dca744f88cb47ab9fac6ba46b1f309efc893","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-06-24 14:23:57.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"2a40f108f65b392f594c97232a8dfad8553e5737","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-24 14:25:02.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/41725cc69aff4dadaa2334f8b65f10bb : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2ae5baea128c4541a0d4ba4e386d1b2a : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"eba6a842c2b7c228f942a1f037581b9ac8c709a2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 15:11:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1036/.","accounts_in_message":[],"_revision_number":4},{"id":"5c10506db6395309bcdc296644021e21bd5be56d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 15:22:10.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1165/.","accounts_in_message":[],"_revision_number":4},{"id":"6236b35ff89c1a9834580d4004cd9989d23913ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 15:39:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b49d6b09cb14293bab6d007d1b3978d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c3b47e1b2fab4c4a8ffd6c14bc2069c2 : SUCCESS in 23m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ca6d65e73a97425ab8ed9e08fd86be1a : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c07c97b4c15944a781a8ad157b650830 : SUCCESS in 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"265368242fb01065a61cd3a707d5b3f9dc3163d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 16:36:25.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfdb0c36f2574bdc9a53101a02b1a9ba\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d6106c8f8e274630b53c823109d5dc35 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d15135451174833a5ee95d7d318fdd3 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bad5cf7cc25d412e8ee8a52c4ad1186f : SUCCESS in 10m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/664b35443c5a4a288cf32ba4cf434b1a : SUCCESS in 7m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/475203c8a1224b7ba0d6b8bb09d205b1 : SUCCESS in 14m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0adfe310e0694a92a8db58ee685c709b : SUCCESS in 12m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa2c60d20ece41aa9e3b90c9ec99e9d5 : SUCCESS in 4m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4992aa7607fe419e97e236279f047df1 : SUCCESS in 59m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdadd32f3bd14ae59e3f99dad3cb93c9 : SUCCESS in 19m 16s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8dce02789b1c46c2a83a499a8816455d : SUCCESS in 5m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dc1b23cdf26340aebe12ac78a41addbf : SUCCESS in 6m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/14537a8ec34c47d28560511c155dde7e : SUCCESS in 9m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d71ea79cb71e492e92feefe05947d516 : FAILURE in 19m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f3fb661d165e44ca9a12670903e1172f : SUCCESS in 5m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2cef62f64c3d47a1bbccf88013b30f44 : SUCCESS in 2h 06m 58s","accounts_in_message":[],"_revision_number":8},{"id":"83e41386fe4388d4f31806013d0e2aba3abf3758","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 17:07:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1038/.","accounts_in_message":[],"_revision_number":5},{"id":"046860d7d5c1c08d0d05bc0e1c4a43292790b1eb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 17:20:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1167/.","accounts_in_message":[],"_revision_number":5},{"id":"9a436b3f6ac9033cca003f8a6b12fd59e179e62b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 18:01:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1039/.","accounts_in_message":[],"_revision_number":6},{"id":"60c447f2b64650b7600a5c8c14839d8e1c73c0d1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 18:17:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1168/.","accounts_in_message":[],"_revision_number":6},{"id":"64921a88ce102414aea8941c469e5d65e6b69275","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 18:54:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1040/.","accounts_in_message":[],"_revision_number":7},{"id":"41e1bf9da54ac3780541e8e9db6da0aa7afd8404","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 19:12:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1169/.","accounts_in_message":[],"_revision_number":7},{"id":"c44926a1457657e77f3d970fb1df4f97a86c785c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 19:49:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1041/.","accounts_in_message":[],"_revision_number":8},{"id":"8fcaab197f56482317c9d3bd88d21d10eb704131","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 20:08:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1170/.","accounts_in_message":[],"_revision_number":8},{"id":"9f9bf2264cbdebe4a8626443f4bcac7249491a24","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-26 09:10:30.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/57/953157/8/20250626134200/NEC-V-ISCSI : SUCCESS in 4h 22m 31s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/57/953157/8/20250626134200/NEC-V-FC : SUCCESS in 4h 28m 27s","accounts_in_message":[],"_revision_number":8},{"id":"c0fba11592a012d2428991f25ae8b3b023007d85","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-27 18:25:39.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/57/953157/8/20250627210825/NEC-ISCSI : SUCCESS in 5h 47m 47s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/57/953157/8/20250627210825/NEC-FC : SUCCESS in 6h 17m 11s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"715b2695d7994aa741a9c52b7045e7fb002159ad","revisions":{"d642a9c2ca8fdd1621b823e074a88aeee63419a3":{"kind":"REWORK","_number":1,"created":"2025-06-23 22:53:33.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/57/953157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/953157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/953157/1"}}},"commit":{"parents":[{"commit":"673adaee29b60e3bc579e797bf2da4e01d177ced","subject":"Fix flake8 warnings on flake8-import-order\u003e0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/673adaee29b60e3bc579e797bf2da4e01d177ced"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-23 22:48:16.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-23 22:48:16.000000000","tz":-180},"subject":"Revert \"Pin flake8-import-order\u003c0.19.0\"","message":"Revert \"Pin flake8-import-order\u003c0.19.0\"\n\nThis reverts commit Ic99814a61c93a9249ae9fbe5ecd5c510cb6e31ed.\n\nReason for revert: The new style checks preventing Cinder\u0027s code\nfrom passing in flake8 have been addressed.\n\nChange-Id: I43d1846c3cacc6bd84655d56f29077f1a642baac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d642a9c2ca8fdd1621b823e074a88aeee63419a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d642a9c2ca8fdd1621b823e074a88aeee63419a3"}]},"branch":"refs/heads/master"},"3740a7141265c75bc696a951ae13e93238f800fa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-23 22:54:12.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/57/953157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/953157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/953157/2"}}},"commit":{"parents":[{"commit":"673adaee29b60e3bc579e797bf2da4e01d177ced","subject":"Fix flake8 warnings on flake8-import-order\u003e0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/673adaee29b60e3bc579e797bf2da4e01d177ced"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-23 22:48:16.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-23 22:54:12.000000000","tz":0},"subject":"Revert \"Pin flake8-import-order\u003c0.19.0\"","message":"Revert \"Pin flake8-import-order\u003c0.19.0\"\n\nThis reverts commit Ic99814a61c93a9249ae9fbe5ecd5c510cb6e31ed.\n\nReason for revert: The new style checks preventing Cinder\u0027s code\nfrom passing in flake8 have been addressed in commit\nI3a442308e031e0c206c82d34ee4516ac6cc21656.\n\nChange-Id: I43d1846c3cacc6bd84655d56f29077f1a642baac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3740a7141265c75bc696a951ae13e93238f800fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3740a7141265c75bc696a951ae13e93238f800fa"}]},"branch":"refs/heads/master","description":"Edit commit message"},"96574d61693f8b1a6de0ef1e82eedd5eb6347b3f":{"kind":"NO_CHANGE","_number":3,"created":"2025-06-23 22:56:55.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/57/953157/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/953157/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/953157/3"}}},"commit":{"parents":[{"commit":"52312affc3a26e7abda7c4169b3eae5efe3ca385","subject":"Fix flake8 warnings on flake8-import-order\u003e0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52312affc3a26e7abda7c4169b3eae5efe3ca385"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-23 22:48:16.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-23 22:56:41.000000000","tz":-180},"subject":"Revert \"Pin flake8-import-order\u003c0.19.0\"","message":"Revert \"Pin flake8-import-order\u003c0.19.0\"\n\nThis reverts commit Ic99814a61c93a9249ae9fbe5ecd5c510cb6e31ed.\n\nReason for revert: The new style checks preventing Cinder\u0027s code\nfrom passing in flake8 have been addressed in commit\nI3a442308e031e0c206c82d34ee4516ac6cc21656.\n\nChange-Id: I43d1846c3cacc6bd84655d56f29077f1a642baac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96574d61693f8b1a6de0ef1e82eedd5eb6347b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96574d61693f8b1a6de0ef1e82eedd5eb6347b3f"}]},"branch":"refs/heads/master"},"769adb8a53873f262fde00309f9818c9c94b592e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-06-24 13:13:54.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/57/953157/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/953157/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/953157/4"}}},"commit":{"parents":[{"commit":"3677f66bd64086bf12803b5073455af0c31a35df","subject":"Fix flake8 warnings on flake8-import-order\u003e0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3677f66bd64086bf12803b5073455af0c31a35df"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-23 22:48:16.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-24 13:13:39.000000000","tz":-180},"subject":"Revert \"Pin flake8-import-order\u003c0.19.0\"","message":"Revert \"Pin flake8-import-order\u003c0.19.0\"\n\nThis reverts commit Ic99814a61c93a9249ae9fbe5ecd5c510cb6e31ed.\n\nReason for revert: The new style checks preventing Cinder\u0027s code\nfrom passing in flake8 have been addressed in commit\nI3a442308e031e0c206c82d34ee4516ac6cc21656.\n\nChange-Id: I43d1846c3cacc6bd84655d56f29077f1a642baac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/769adb8a53873f262fde00309f9818c9c94b592e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/769adb8a53873f262fde00309f9818c9c94b592e"}]},"branch":"refs/heads/master"},"db849c3cfd8712420f5692518209ccea746517c3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-06-24 14:02:09.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/57/953157/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/953157/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/953157/5"}}},"commit":{"parents":[{"commit":"781668f4451a3089fbbb313bbb6829875aa2b2a6","subject":"Fix flake8 warnings on flake8-import-order\u003e0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/781668f4451a3089fbbb313bbb6829875aa2b2a6"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-23 22:48:16.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-24 14:01:52.000000000","tz":-180},"subject":"Revert \"Pin flake8-import-order\u003c0.19.0\"","message":"Revert \"Pin flake8-import-order\u003c0.19.0\"\n\nThis reverts commit Ic99814a61c93a9249ae9fbe5ecd5c510cb6e31ed.\n\nReason for revert: The new style checks preventing Cinder\u0027s code\nfrom passing in flake8 have been addressed in commit\nI3a442308e031e0c206c82d34ee4516ac6cc21656.\n\nChange-Id: I43d1846c3cacc6bd84655d56f29077f1a642baac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db849c3cfd8712420f5692518209ccea746517c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db849c3cfd8712420f5692518209ccea746517c3"}]},"branch":"refs/heads/master"},"e5e9d94fa06acad5cc9201f422febec04a5ed8f6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-06-24 14:20:41.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/57/953157/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/953157/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/953157/6"}}},"commit":{"parents":[{"commit":"4b96a9a88e350b6c1192f192b07c9af1f0f31747","subject":"Pin flake8-import-order\u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b96a9a88e350b6c1192f192b07c9af1f0f31747"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-23 22:48:16.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-24 14:20:41.000000000","tz":0},"subject":"Revert \"Pin flake8-import-order\u003c0.19.0\"","message":"Revert \"Pin flake8-import-order\u003c0.19.0\"\n\nThis reverts commit Ic99814a61c93a9249ae9fbe5ecd5c510cb6e31ed.\n\nReason for revert: The new style checks preventing Cinder\u0027s code\nfrom passing in flake8 have been addressed in commit\nI3a442308e031e0c206c82d34ee4516ac6cc21656.\n\nChange-Id: I43d1846c3cacc6bd84655d56f29077f1a642baac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e5e9d94fa06acad5cc9201f422febec04a5ed8f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e5e9d94fa06acad5cc9201f422febec04a5ed8f6"}]},"branch":"refs/heads/master","description":"Rebase"},"09ba3e058bd346f4ff2da5bcbbad4b97df272cb7":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-06-24 14:22:45.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/57/953157/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/953157/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/953157/7"}}},"commit":{"parents":[{"commit":"4b96a9a88e350b6c1192f192b07c9af1f0f31747","subject":"Pin flake8-import-order\u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b96a9a88e350b6c1192f192b07c9af1f0f31747"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-23 22:48:16.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-24 14:22:45.000000000","tz":0},"subject":"Revert \"Pin flake8-import-order\u003c0.19.0\"","message":"Revert \"Pin flake8-import-order\u003c0.19.0\"\n\nThis reverts commit Ic99814a61c93a9249ae9fbe5ecd5c510cb6e31ed.\n\nReason for revert: The new style checks in 0.19.x are preventing Cinder\u0027s code from passing in flake8, and have been addressed\nin commit I3a442308e031e0c206c82d34ee4516ac6cc21656.\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/953156\n\nChange-Id: I43d1846c3cacc6bd84655d56f29077f1a642baac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09ba3e058bd346f4ff2da5bcbbad4b97df272cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09ba3e058bd346f4ff2da5bcbbad4b97df272cb7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"715b2695d7994aa741a9c52b7045e7fb002159ad":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-06-24 14:23:57.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/57/953157/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/953157/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/953157/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/953157/8"}}},"commit":{"parents":[{"commit":"4b96a9a88e350b6c1192f192b07c9af1f0f31747","subject":"Pin flake8-import-order\u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b96a9a88e350b6c1192f192b07c9af1f0f31747"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-23 22:48:16.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-06-24 14:23:57.000000000","tz":0},"subject":"Revert \"Pin flake8-import-order\u003c0.19.0\"","message":"Revert \"Pin flake8-import-order\u003c0.19.0\"\n\nThis reverts commit Ic99814a61c93a9249ae9fbe5ecd5c510cb6e31ed.\n\nReason for revert: The new style checks in 0.19.x are preventing Cinder\u0027s code from passing in flake8, and have been addressed\nin commit I3a442308e031e0c206c82d34ee4516ac6cc21656.\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/953156\nChange-Id: I43d1846c3cacc6bd84655d56f29077f1a642baac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/715b2695d7994aa741a9c52b7045e7fb002159ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/715b2695d7994aa741a9c52b7045e7fb002159ad"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
