)]}'
{"id":"openstack%2Fcinder~953238","triplet_id":"openstack%2Fcinder~master~Ic8a63bd38b9868898167ab066e11147461ffd684","project":"openstack/cinder","branch":"master","topic":"repl_interval","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2025-06-24 19:42:31.000000000","reason":"\u003cGERRIT_ACCOUNT_13425\u003e replied on the change","reason_account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},"35316":{"account":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"last_update":"2025-08-06 19:44:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8a63bd38b9868898167ab066e11147461ffd684","subject":"[Pure Storage] Fix snapshot replication interval issue","status":"MERGED","created":"2025-06-24 19:01:45.000000000","updated":"2025-08-07 13:12:08.000000000","submitted":"2025-08-06 19:44:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"953238-repl_interval","meta_rev_id":"11b276068c03dbf8257bb5cc3e324dc9d6b85ed0","_number":953238,"virtual_id_number":953238,"owner":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 19:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"recommended":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":1,"date":"2025-06-24 19:42:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2025-08-06 17:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-16 19:57:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2025-08-06 17:33:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}],"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-24 19:02:56.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-24 19:33:33.000000000","updated_by":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2025-06-24 20:26:04.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-06-24 20:51:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-24 20:55:47.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-06-24 22:09:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-27 08:18:58.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-06-28 00:41:34.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-07 14:17:35.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-16 19:57:04.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2025-08-06 17:00:17.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-08-06 17:34:12.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"fcc7a1829fd30ae7905a0a9f84b01e25901559b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2025-06-24 19:01:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"102fbf864aa468267a57e577d652ec75a06c30da","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-24 19:02:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/375df7709d9d4658ae33aa6c745b01ed : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d33f1b4fa15f46a6928d4856f6653085 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"533400dd2b927196b02cd61b12d58c4ec8cd07a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2025-06-24 19:31:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf041c987643abe9c6260b24628b3d6d0b603579","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-24 19:32:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3a9c50c707a34f3cbcf029a5c5553aed : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6317ed87cb304b4d95b3d39e8a276c18 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"9e35df71bc8bc8f79dd38c8a48c63b14798cdff1","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-06-24 19:42:31.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"972b9a73aa657a0208896ac2e75c7cb5273c203c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-24 20:26:04.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/90e413359d704965a48fe70ddafcf97d\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d9d3ac2afed449d891c77f236178e909 : NODE_FAILURE Node request 200-0000068910 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/80c371a3401f4df182def4b631dca8b4 : NODE_FAILURE Node request 200-0000068911 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ec2fdb19d5bc482e87ca5b603d94203f : SUCCESS in 53m 36s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/19c28afda59f4caa986b93b369574fb7 : RETRY_LIMIT in 24s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f7ea256314614171bdc5f8ff7bd9f6b5 : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":2},{"id":"096d9da92c5a8f96d72daf7cfed8d0c066c879de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 20:51:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b087585eea14ced9eb1eac2dbaeae7f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/836dde5f4f82455597bc7b54678f75ee : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c2c0a8fa5fa4f36a49b422c5cc49938 : SUCCESS in 24m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/489690d0001c4822b13085dc0be2c31a : SUCCESS in 25m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b24eab4f853808b57043ac18fa11a09d8b8852be","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 20:55:47.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1042/.","accounts_in_message":[],"_revision_number":1},{"id":"035a53106b3e886d664b98c6ed829c7b840f0e46","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-24 21:13:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1171/.","accounts_in_message":[],"_revision_number":1},{"id":"c8fad1160052d25261deb5d193cf4cc36a150b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 22:09:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e501ca62e8b438a802ef2dabc8d9998\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66cf73cea62640669c66e9b0344a6c3c : SUCCESS in 6m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c59c5dd483bf42569dfba8091a341e73 : SUCCESS in 9m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25654c38d32b4a1da5c69b56e6c5270c : SUCCESS in 11m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59c4a356b2d84df8920bea51d7fbe2ee : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da62f890781b417d9bb26d100504b0e5 : SUCCESS in 15m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/c4ec42afb19b4d3ab670afc06b323594 : FAILURE in 8m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1a588d6a6314745bceea9dfb9c9fc77 : FAILURE in 8m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7625a9c9d754f3db6dac2f4bfa71233 : FAILURE in 8m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c2e69f8ea5b4ec695b2610808abfe1c : SUCCESS in 47m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/672152f3b02f4cd2b9a788995832f212 : SUCCESS in 23m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e96fd914e784753a02aa2355582b0f6 : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/18ed26e76ec74fbd9d24ebd2c67a1bba : SUCCESS in 3m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de947743de534624b0063d7920990629 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1e1fb6b061649a789c37d60b03aee84 : SUCCESS in 3m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/eba6b3b70a7a468aae0eb95a4df8b170 : SUCCESS in 5m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/609a66b688fb482f8a368cf4b7416d85 : FAILURE in 34m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e51aac18f71e4b028eedcb40aba8cec8 : SUCCESS in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c6968394922847ec86dc86ceb136b040 : SUCCESS in 2h 06m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fc03a93fab81459ea83fe1adf74a69ed : FAILURE in 2h 35m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/779e756f31de4769b4dbae6230881e12 : SUCCESS in 2h 10m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7736acd4442e4a4b976c57fa62b23cdc : FAILURE in 17m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c0c6b875f1904c96a6905d1b62b5e324 : SUCCESS in 38m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e18344a812b04ec7844d702831a42ea2 : SUCCESS in 33m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d416f7d368624a8cbc4da5a820090340 : SUCCESS in 1h 35m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3387a5bb2bd34693bf7812c83dc40965 : SUCCESS in 22m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e1ab8a499bdb430abe84049291a775ab : SUCCESS in 1h 30m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0229ce7c72cc4db79013918635aa7ed1 : FAILURE in 9m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e1ae1149d22e4b048e55dc36dcb3f8ac : SUCCESS in 1h 52m 03s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/492f3958a93b48ffbbce5c5dae3f5872 : SUCCESS in 1h 24m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7656d2c45214a65afd03e2c47798a7d : SUCCESS in 30m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f05a774c77b509aeb4ff8037efcf6eeb52d64d91","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-25 02:57:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1048/.","accounts_in_message":[],"_revision_number":2},{"id":"5e373d55ada21b4d928bb42f99ff9d3b3a664d3e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-25 03:03:26.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1177/.","accounts_in_message":[],"_revision_number":2},{"id":"703e0db9f1b19e8740ec611679faf7028f06742a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-06-27 08:18:58.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/38/953238/2/20250627042554/NEC-V-ISCSI : FAILURE in 10h 40m 25s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/38/953238/2/20250627042554/NEC-V-FC : FAILURE in 12h 53m 01s","accounts_in_message":[],"_revision_number":2},{"id":"88411d4303e3f6bd96e6b000f87f0c86e498ed48","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-06-28 00:41:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/38/953238/2/20250628032542/NEC-ISCSI : SUCCESS in 5h 49m 46s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/38/953238/2/20250628032542/NEC-FC : SUCCESS in 6h 15m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1a681dd75d7d46cbfde0fdd4a12ad84b75c765f7","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2025-07-07 08:55:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d73f6ac3f8d49ab73214f8d00d46d96bad26121","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-07 08:56:34.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3917f98faa7ee17a483ce9cad1650d2ff838b9af","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-07 09:53:38.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a7d4b3278e1541728a04311bcb080139\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5a910c1ffbc74dbf96d432e982627081 : NODE_FAILURE Node request 200-0000071339 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/31d846b4e38f402e8a75200fe96ae3ea : NODE_FAILURE Node request 200-0000071340 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8e727dab5a73405cab60e528a614cb8c : RETRY_LIMIT in 24s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6570663defab47c1bc107a5a6f1c47dc : NODE_FAILURE Node request 200-0000071372 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fa77291ddb644f5ea7382fcedd148e8b : RETRY_LIMIT in 26s","accounts_in_message":[],"_revision_number":2},{"id":"e608cc5140f0fa42b3ebe8fb07b67fdd28976cbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 11:04:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14374ba3083248ee8063e97fbb972fe8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4275fedcf0894c7f9951d69e09503328 : SUCCESS in 29m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d926fe0c60504d5691c021eec70bd91a : FAILURE in 32m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1b684bc3800f4c57b184edeb37e6e353 : SUCCESS in 47m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a694cccb881e057695bb55a87bc3d424be1f191a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 11:16:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dcb58bf305640ddb5d82f4cf86f0d3f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c154ec6dad64f31a6a939adfea08cc9 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0b2eab87c774b768d915caf7e2c1e1b : SUCCESS in 7m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01ef6fcc0cb94f3690e08ec19f4c31c4 : SUCCESS in 11m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/493e8de425ae4589ba995579ecb25e26 : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb74ebf103264395b5fdc368749b6f80 : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/d112a8e37539450dbdb0779dc9ac7f1c : SUCCESS in 55m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7bc7d13017045e5b1e7d699a794c3dc : SUCCESS in 33m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e8b3b41d8ca4ce1ad745c463d5c54be : SUCCESS in 1h 46m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c927cc9efdf4a20a40ce63918fc0267 : SUCCESS in 1h 03m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82fb171b8cd34b7aaf742796edae8f98 : SUCCESS in 21m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ed5864e63cdc47d4bc8f1bb1947e7002 : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2be01ed5ab614d58a676e534ef660045 : SUCCESS in 6m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2729f00af2d04d3ba3c80ec857c0f687 : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/631b355235a34215800711fda48a9696 : SUCCESS in 3m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c064db113f3849c2a6e31e82ea7699c0 : SUCCESS in 9m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b7abc614652c4ea2a112e86854b40a35 : FAILURE in 34m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/305b6ea86bfd422eb89d5ef606d7f2f8 : SUCCESS in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d35d8fb149554cbd818ccc8a4066a307 : SUCCESS in 2h 05m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/209a305a4ec548d4b78fb5aaa7a3e387 : SUCCESS in 1h 26m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cc34efff1c024fff9e86091250e3e9a1 : SUCCESS in 2h 14m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8ec5b48b9c85412b8abcdde18dadacc1 : FAILURE in 18m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d55cc57f6c364d989bc57f41705c8ddd : SUCCESS in 29m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/15bc45e12cf04b4e86c5c4560192143f : SUCCESS in 1h 05m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/09deb63975a04f8da6f2e9dd71e35056 : SUCCESS in 1h 23m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c0b61e55a0794300aba721129d7756a6 : SUCCESS in 42m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d7c5ac2f0f364e26bc508734d79f3a7f : SUCCESS in 1h 34m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b93d4a1b77a34dde9f682bf3bfdb3691 : FAILURE in 12m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a37e3765aa04cec943fb8af10d3a838 : SUCCESS in 53m 39s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9397ed5c379b4b28a6662f235ac30715 : SUCCESS in 1h 35m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2601acd5e198479fb28fbb095e3d435e : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":2},{"id":"91b8df64c451b19fe1a0522ea37f3bdc4413887b","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2025-07-07 11:52:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5fc833bbe5d94aeb270bfd7517f992371ad4dd39","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-07 14:17:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"281e8f4fc25614360efd34015c120ad9acb8e801","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-07 14:43:47.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4d78980c23b94346800062a57a4df62a\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/74ec914390024455a7d5c6f4ebe5ca1f : NODE_FAILURE Node request 200-0000071388 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dcf2f1b652384951b62438767bbf7377 : NODE_FAILURE Node request 200-0000071389 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/23b6c3e4567d40f9a9350730201d987b : RETRY_LIMIT Host unreachable in 48m 44s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3cc307c63ac74821ae23e4437b6d5395 : RETRY_LIMIT Host unreachable in 1h 08m 52s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8c6e5dd2b301432dbb054d751135c1b4 : RETRY_LIMIT Host unreachable in 44m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c9f5d118b27ac67d5fa1c1902d4758940d31da0f","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2025-07-08 06:18:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee279024036c1002c9b041065fdd6eafe3b1f34d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-08 08:02:53.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a9659adc0ecb463998cd926db3e42fd4\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/94901489ccb9462ab583b40198c7cb09 : NODE_FAILURE Node request 200-0000071469 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c41b34f4b493430581e232ac7bb77037 : NODE_FAILURE Node request 200-0000071470 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/03ce744101cb4a44bba066ff71baba69 : RETRY_LIMIT in 26s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/afc1b2433d2c4d6e8675f2223cf80e00 : NODE_FAILURE Node request 200-0000071525 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/993ae95283464d89afe88b96c60e598c : RETRY_LIMIT in 26s","accounts_in_message":[],"_revision_number":2},{"id":"b1772cfb97140817b5af94d9cf6dd7c79202c9c9","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2025-07-09 06:26:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2e1f2d77caed49a5415f73e95a2acada4c1e2e7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-09 09:02:39.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2ec38a664b9f449faf46a190a3b00505\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bdb9fdc8f66540b382d4fd2486be4f08 : NODE_FAILURE Node request 200-0000072413 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/93d1a2205aa1466c95dd96ae681fc99b : NODE_FAILURE Node request 200-0000072414 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ecc3343255c14fa1beb836e7e88d4edf : RETRY_LIMIT Host unreachable in 49m 50s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b326d0ab0c404724938ba0aa075e130f : RETRY_LIMIT in 25s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7c4991213c3b40018961a824a5591d0d : RETRY_LIMIT Host unreachable in 47m 38s","accounts_in_message":[],"_revision_number":2},{"id":"03fadc381d26a4227f88cfcd9e7387f962298a79","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2025-07-11 06:55:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"753f45a5dfbce0e7eede2a6e1cc0056c653dd0fe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-11 07:48:38.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c01f743cf0594e6788c9458cb217282d\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/111211c0a34b4954b44bb38f157c65f8 : NODE_FAILURE Node request 200-0000073642 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8fba338909f2408d83f61ca991a4d1a1 : NODE_FAILURE Node request 200-0000073643 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/268e10c3553443449bea4da3ee484eb7 : FAILURE in 53m 02s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f31512c5ecad4689bd68f23c1329556d : SUCCESS in 44m 18s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4b4bbd08d1f04895bec4bd955ad55611 : SUCCESS in 48m 08s","accounts_in_message":[],"_revision_number":2},{"id":"87321009cac0095d9d1db4733811fcfba19a662d","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-07-16 13:19:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"067b0ec92e5fe40a8ece7d32e830665c807bb019","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-16 15:17:18.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3ddc0088030e4ca29c0f55e69f137a7a\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1d19c99d24064618a414729482668be0 : NODE_FAILURE Node request 200-0000073885 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3af11dec62f24e6fb3ef3ec6654a371c : NODE_FAILURE Node request 200-0000073886 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bb98512bdcbc4ecdb9a53cdb83d9f42b : SUCCESS in 58m 52s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5a5f6f908d0c4350a8da74dd4b7237d0 : FAILURE in 1h 57m 00s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dc8be55cc839482284d361eba96c1de8 : SUCCESS in 43m 38s","accounts_in_message":[],"_revision_number":2},{"id":"8d3d1826a2dbdd709e27a4d5eb71ddc0a1655c3f","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-07-16 15:38:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89c8b7e9cf8e7e6f40cc16d1fa07f94645736960","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-16 16:36:18.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b2ea5188cfcf494b85bdb8db60b55df3\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ad128a87d7334d2aac2af653a4ccc85c : NODE_FAILURE Node request 200-0000073943 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cfc4ecf4207c431d8b573b47e11f8f94 : NODE_FAILURE Node request 200-0000073944 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0d3243c4a60f412e92243b62394be160 : SUCCESS in 46m 38s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/090bffbd82674c39b1856d51aa12817c : FAILURE in 56m 53s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eaca8c53504d47deae910574e17b8564 : SUCCESS in 42m 39s","accounts_in_message":[],"_revision_number":2},{"id":"da2d94ef30f8532ccae8d4da6259689c54dc3251","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-07-16 19:57:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"036fabb52eb75d991c6a41ed8716e5382b479b77","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-07-31 16:05:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dbe9414af2c8430236db301f2bdd7271da95f20f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-31 17:10:50.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1bf6cb1c92a34e358bc83292a675bf17\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d0ee569b23e44723bd5846a784ef53e6 : NODE_FAILURE Node request 200-0000077002 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e61359e679d649559d623adfb80f3984 : NODE_FAILURE Node request 200-0000077003 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/aa348098390642078227a0ce41787faa : SUCCESS in 52m 21s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d108e56fe26e47d9ba945a14013a15e6 : RETRY_LIMIT in 25s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9e758891a4474ca2b31b543569e385d2 : SUCCESS in 49m 12s","accounts_in_message":[],"_revision_number":2},{"id":"81c0eea9eab6c673d991100542fe6ebabe854156","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-07-31 17:19:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf77c3f88db7d9e8c74cd2d4e9e987a034e3a87e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-31 18:20:51.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e3ec3237aee24b8e9196a3230dc31b02\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d716ec0bd36d4e858ccf02630eb11bb6 : NODE_FAILURE Node request 200-0000077047 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8356589cf3c64232ba146d1ded99e136 : NODE_FAILURE Node request 200-0000077048 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/472a1086d1f545168aa7785100f00dea : RETRY_LIMIT in 26s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/211c30377eae4dcc8a040514e990bfc6 : SUCCESS in 52m 17s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e364cd90206e494f94b6270fb954c2a7 : RETRY_LIMIT in 24s","accounts_in_message":[],"_revision_number":2},{"id":"a521d9ff81a39b982a4472035ea4be6c5bbd644d","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-07-31 18:38:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d86ba582b53dcddc0be545ced39d431885f4513","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-31 21:17:17.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/252d9d64fed54046b0efc70121cca804\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4754dc5d5b0c4c31952efee48ba35a67 : NODE_FAILURE Node request 200-0000077123 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8aee17e883b94a4183874b9f5c603a6d : NODE_FAILURE Node request 200-0000077124 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f89070fc2cca44da814fb567bab0607a : RETRY_LIMIT in 24s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de2e45420ace4082884603b1eeb28d19 : RETRY_LIMIT in 25s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f24f2813921647a6abc8e721f446d535 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":2},{"id":"b07510b3510605d3f60d9faede6568952d8ebed9","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-08-06 17:00:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"706b33bea46bc28f61fb98ed7c107ce7e607f835","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-08-06 17:33:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84dfcdf001b9421b1644ac86a344ac64dadebe43","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-06 17:34:12.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3f4abe0a19bef340ae6a364f63f0df18252d4c4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 17:34:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"488c9bfc07569409bae4316183a2a4ef00d75424","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 19:44:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf87ee45f8aa45539bb12db582013e67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21205de8d82b4db19c4404f55a4d925c : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c56a2ba35f344e8688c0f456ac750d72 : SUCCESS in 11m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/786a836ce35c41c39e2bf254d06ecf71 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a5c89a3fe34fc28ebfee87d2947594 : SUCCESS in 14m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/2fa9a42d3c414aa5bff7bf7f3ca23bfe : SUCCESS in 56m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ca261efbed9142adab58abaa561ce323 : SUCCESS in 1h 03m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/62d08c62ad504dd1b466948e630eff78 : SUCCESS in 1h 21m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ffbb9b6ab38466dbbeb05dd7da442b1 : SUCCESS in 59m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8141a7cc51a24338a183398472c2493a : SUCCESS in 22m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6a0ab1a234e84923a52d699570237d90 : SUCCESS in 1h 11m 26s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c48e062a0b3c49198b97309bcd1a235a : SUCCESS in 2h 04m 35s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1732cf468f84167a5d0884c383dc09c : SUCCESS in 1h 50m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47ded79626a94ffaa99bdb703c741dcb : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":2},{"id":"e18e29a7fee885f62f511193b66321b2e437689f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 19:44:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4a87866e459f074b36d3e64354da24f94543b85c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 19:45:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8614052cc5a455190e3bc5982e947cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3107f618f7b419eba33b8ccb8233add : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ac9cfb1b8eb447982eb8ff8aa8809cb : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2},{"id":"11b276068c03dbf8257bb5cc3e324dc9d6b85ed0","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-07 13:12:08.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/74e1df7e7e0f408fb21447f787223e00\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/0cd1c6c54f504b69873687bc834c9a9d : FAILURE in 8h 11m 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a14e59523191558645477ddfa95c14449a60230e","revisions":{"5df4fe70e41ca45fdb9cd35a7bb7eec2ab5a2751":{"kind":"REWORK","_number":1,"created":"2025-06-24 19:01:45.000000000","uploader":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"ref":"refs/changes/38/953238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/953238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/953238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/953238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/953238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/953238/1"}}},"commit":{"parents":[{"commit":"111e43b9c05c3a7ac8e14771d976bc6c68a70d9c","subject":"Merge \"tgt target: Provide unique scsi_sn and scsi_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/111e43b9c05c3a7ac8e14771d976bc6c68a70d9c"}]}],"author":{"name":"Keerthivasan S","email":"ksuresh@purestorage.com","date":"2025-06-24 18:58:50.000000000","tz":-240},"committer":{"name":"Keerthivasan S","email":"ksuresh@purestorage.com","date":"2025-06-24 18:58:50.000000000","tz":-240},"subject":"[Pure Storage] Fix snapshot replication interval issue","message":"[Pure Storage] Fix snapshot replication interval issue\n\nIn cinder.conf snapshot replication interval can be set in seconds.\nHowever, backend code expects interval in milliseconds.\nAdded fix to handle the conversion to set replication interval properly.\n\nChange-Id: Ic8a63bd38b9868898167ab066e11147461ffd684\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5df4fe70e41ca45fdb9cd35a7bb7eec2ab5a2751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5df4fe70e41ca45fdb9cd35a7bb7eec2ab5a2751"}]},"branch":"refs/heads/master"},"a14e59523191558645477ddfa95c14449a60230e":{"kind":"REWORK","_number":2,"created":"2025-06-24 19:31:56.000000000","uploader":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"ref":"refs/changes/38/953238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/953238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/953238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/953238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/953238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/953238/2"}}},"commit":{"parents":[{"commit":"111e43b9c05c3a7ac8e14771d976bc6c68a70d9c","subject":"Merge \"tgt target: Provide unique scsi_sn and scsi_id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/111e43b9c05c3a7ac8e14771d976bc6c68a70d9c"}]}],"author":{"name":"Keerthivasan S","email":"ksuresh@purestorage.com","date":"2025-06-24 18:58:50.000000000","tz":-240},"committer":{"name":"Keerthivasan S","email":"ksuresh@purestorage.com","date":"2025-06-24 19:30:26.000000000","tz":-240},"subject":"[Pure Storage] Fix snapshot replication interval issue","message":"[Pure Storage] Fix snapshot replication interval issue\n\nIn cinder.conf snapshot replication interval can be set in seconds.\nHowever, backend code expects interval in milliseconds.\nAdded fix to handle the conversion to set replication interval properly.\n\nCloses-Bug: #2115284\nChange-Id: Ic8a63bd38b9868898167ab066e11147461ffd684\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a14e59523191558645477ddfa95c14449a60230e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a14e59523191558645477ddfa95c14449a60230e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
