)]}'
{"id":"openstack%2Fcinder~953958","triplet_id":"openstack%2Fcinder~stable%2F2024.1~I5da7ae04b87b4fd52a682a6545060e852174f6c8","project":"openstack/cinder","branch":"stable/2024.1","topic":"storpool-backports-2024.1","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-07-04 07:22:58.000000000","reason":"Change was marked work in progress"},"35429":{"account":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"last_update":"2025-10-22 15:32:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5da7ae04b87b4fd52a682a6545060e852174f6c8","subject":"StorPool: drop _attach_volume() and _detach_volume()","status":"ABANDONED","created":"2025-07-02 14:01:50.000000000","updated":"2025-10-22 15:32:15.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"cherry_pick_of_change":787296,"cherry_pick_of_patch_set":22,"meta_rev_id":"650042063727fc7fcd6dd4bc3ec40a671f6dc8c1","_number":953958,"virtual_id_number":953958,"owner":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-04 09:46:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-2,"date":"2025-10-22 15:32:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-07-02 14:01:50.000000000","updated_by":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"reviewer":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"state":"CC"},{"updated":"2025-07-02 14:56:14.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-07-02 16:04:53.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-07-02 18:46:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-02 18:51:20.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-07-02 21:47:15.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-07-02 21:48:12.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-04 08:00:02.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"fdb09bcbfb79666b129798696e9813aee26b7acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-07-02 14:01:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26b2931d92dc7c24fa020bf31d9aaba0062afab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-07-02 14:20:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fecb8977111157a82b6577a818148b867ec08c96","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-02 14:56:14.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1131/.","accounts_in_message":[],"_revision_number":2},{"id":"ef9ebd97d9a26f8e6bb8818f9cd46a5f3febdf08","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-02 14:59:17.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1260/.","accounts_in_message":[],"_revision_number":2},{"id":"0a1294a540dfac25818f3c2af6ef9fca2796d48b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-02 16:04:53.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cd46a4d9ce0a4a209fa0038e8140a7b8\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/323f17382ee941a9835340407d0f5584 : RETRY_LIMIT Host unreachable in 12m 12s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6357e2c99971420ba4a4205aed6d57e4 : NODE_FAILURE Node request 200-0000070099 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1f93b29fe8914e9ebc5f4eb454360fc3 : RETRY_LIMIT Host unreachable in 1h 01m 57s","accounts_in_message":[],"_revision_number":2},{"id":"d0df6933d446718f5298bbeb4cc9151507dbd8dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 18:46:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e9c3f1030f146b4b1f3792a68ca99a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5922a1d532e4fc78949a68773bc3a18 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db92be265ce6457c98d34d3e6b929780 : SUCCESS in 10m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ecf9e75d84049099575174ca69e0501 : SUCCESS in 12m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/541d7576eb6f4421b56b8623b1ef68bf : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/da98526758cc4a05aa43dbdb3c2bb02d : SUCCESS in 1h 02m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6da06dadd87f4bbba2f1888e090a82f4 : SUCCESS in 25m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/521954dcdac24840b9fc66fdb93247e0 : SUCCESS in 1h 26m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd49d2a3ac4243e5bea406445cb11a5a : SUCCESS in 53m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6258a1f13a4e4734b39e540bc81b276d : SUCCESS in 22m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/53e15c7b5cfb4a8b9f51405eea03c82c : SUCCESS in 19m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/476ecd63b0a84486a3b38a9e9c39e3d1 : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/621476e0fc5c466ea326432dbcedfb2e : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8c950a57e53a43438c52f4ecb45dcab0 : SUCCESS in 3m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/70decbab8df041cebeac71ead51a8f96 : SUCCESS in 3m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5ff5e679f402454c9a6a86247c119a78 : SUCCESS in 37m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/956d4dd8c4d04e909ea64c03a0f74ed8 : SUCCESS in 5m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/87b37742095742a2a5246f733f34a768 : SUCCESS in 1h 29m 15s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc76102e701042d5818f43df637d97f6 : SUCCESS in 1h 58m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/47b82f39220d485fa07d9d7f1854a477 : SUCCESS in 2h 01m 52s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/efc9435ad29743209099fb7d0c7b648d : SUCCESS in 1h 10m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/436bdc93c0bb45ec9ff1ce4dd4b0e5e8 : SUCCESS in 1h 23m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2e32b4c032ac4ca49cac92a269a71e18 : SUCCESS in 24m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4ae3f9d3a11b4b5db20f4cda2ded4b47 : SUCCESS in 1h 01m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/887516c2d1404909ad335e1450b21fb5 : POST_FAILURE in 2h 28m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/587ad7907f35445caf73ae12501e24c3 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":2},{"id":"d77ec64b4307bc0e7d3843ceb141758d3dc6f66d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-02 18:51:20.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_07_2025_18_46_46\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"03224ca46b314a43f3148e758d4473645c103b43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 20:45:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce6ce2fcbef748d588f0e38f50f8d2a7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad79883c0df34028a634cf254d3eea6f : SUCCESS in 19m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2128292694d54444add0a65b4e92e065 : SUCCESS in 24m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d5214828a653a5d5e57ca4fe16c145e0cf29c70","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-02 21:47:15.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"efbb690e04378723a622b74f6206a418b92846e3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-02 21:48:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"821c7dba0c7ef2c12d514422ce6bc493c8005bd7","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-07-03 11:15:25.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"eb38243a71a62b49e95761214e6e1f98b62ddbef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-07-04 07:22:26.000000000","message":"Patch Set 3: Cherry Picked from branch master.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"31159ce6b72d70263067eaf58f1d61f0e4af71e9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-07-04 07:22:46.000000000","message":"Topic storpool-driver-fixes-stable/2024.1 removed","accounts_in_message":[],"_revision_number":3},{"id":"2109b6b71c8930a67205fc752f64448d835a709b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-07-04 07:22:58.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"8104995ba868c3d9c0ea8ec432ac5313db1edc24","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-04 07:30:32.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1269/.","accounts_in_message":[],"_revision_number":3},{"id":"60e2b2aeb4df11040b3d51356380158a3ced1eee","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-04 07:38:56.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1139/.","accounts_in_message":[],"_revision_number":3},{"id":"4568ca20a0a53db1d1fff4019e565e97119ec594","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-07-04 07:38:58.000000000","message":"Topic set to storpool-backports-2024.1","accounts_in_message":[],"_revision_number":3},{"id":"7ff7244cbddbaac80ca0eaf4b8ecae0ddd1fdb30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 07:41:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb6a5c870b74468ca9148fbf660f6652\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d29554f1c8c64ed58365a1a603427186 : NODE_FAILURE Node(set) request 300-0027359129 failed in 0s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2e31b0ffac024791bae692127937ba05 : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"44f99a63dbd715531bf068b0a0d5f4844ea50861","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-04 08:00:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/953958 : FAILURE in 35m 55s","accounts_in_message":[],"_revision_number":3},{"id":"98819ed3bb920744fc8eacb0ca81445f00a04dde","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-04 08:41:18.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c43f6b4dd62d4112835534dd0d6d0c6f\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/817276012db24ae5b5056fbf4ed37260 : RETRY_LIMIT Host unreachable in 11m 00s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/38a35dfc6b244dedbba46e8474cb4daa : NODE_FAILURE Node request 200-0000070419 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/218511966ba44345af836e32e8eda5b3 : RETRY_LIMIT Host unreachable in 11m 05s","accounts_in_message":[],"_revision_number":3},{"id":"b630c1a48dea9003e1d21a50782264c43e42c3a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 09:46:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c50bf5271ac4d4d831db18e011b11d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24ae8208b397458fb66fe2d4eb362b91 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d1b18bbf70541cfb4ac7f8b05c66c18 : SUCCESS in 15m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca8c3af3d7d744f08211135172577644 : SUCCESS in 12m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921915bd6fbf4fd280d0bff9ee13ec4c : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/21897f4f9c5e400ea92997f8ed3c2103 : SUCCESS in 1h 24m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/95c9a5f055a74462b6d48127b6a98ce5 : SUCCESS in 54m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c7db255d4c88419e9d633453a4052346 : SUCCESS in 1h 28m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63b4c49f11544a759d8c6545726632d6 : SUCCESS in 56m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f2914ee2d4d40d69b0673e1a016c575 : SUCCESS in 28m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/683c80c771c544cba76c49dac1b926d1 : SUCCESS in 13m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3f896c12950148f8af8b97d9dc33b6a5 : SUCCESS in 4m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/df110a2b7dda4397bec72a0e06409892 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac3065d68f454e93a8aa55db0d3c4d1d : SUCCESS in 4m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/81b39830e14d45de80ca145dc19d6a0c : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/417caf23009042abad4882b36ccde61e : SUCCESS in 30m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9449f3be98254c78bb967185d3f15c35 : SUCCESS in 5m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d0b86f2333494eeeadf84943a5a27ade : SUCCESS in 1h 51m 21s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ffed72863d9b4793a6342cf32f28613c : SUCCESS in 1h 41m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e3ccdf9881f2402d9ef611fb9a300519 : SUCCESS in 2h 07m 05s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e153546f56214acdaafbc594cbeece83 : SUCCESS in 53m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b2a79d8db47f495ab932a9dd2515db44 : SUCCESS in 1h 19m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c4d0e634d8f44a82b0554bccffdc296a : SUCCESS in 37m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/edd0905ece6d44ac9ce1c97e00e6f231 : SUCCESS in 1h 24m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/66aa31ee819243a6b0c5ceae5079d386 : POST_FAILURE in 2h 17m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da69068aeb7141f3bda1093437c20a62 : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":3},{"id":"07edd643d539c04d590675ad7307027004479b1a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-04 09:51:34.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_07_2025_09_46_51\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"ce712692e1e81892bc8b201465efe7f64d9d0d4d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-04 13:00:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"eace905ca8721087068dd104f0c0a6ac0e03736f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-22 15:32:03.000000000","message":"Patch Set 3: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"650042063727fc7fcd6dd4bc3ec40a671f6dc8c1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-22 15:32:15.000000000","message":"Abandoned\n\nbranch being deleted","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"27f168a9889b368a9fab649a705f71e4e4e02282","revisions":{"280335b901045b3fa94b3517db363b0e88290123":{"kind":"REWORK","_number":1,"created":"2025-07-02 14:01:50.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/58/953958/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/953958/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/953958/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/953958/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/953958/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/953958/1"}}},"commit":{"parents":[{"commit":"0b50390766d19f15aacf82a91c36ce7c830dfc71","subject":"Pin flake8-import-order\u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b50390766d19f15aacf82a91c36ce7c830dfc71"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2021-04-20 14:46:41.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-07-02 14:01:38.000000000","tz":180},"subject":"StorPool: drop _attach_volume() and _detach_volume()","message":"StorPool: drop _attach_volume() and _detach_volume()\n\nOur os-brick connector already handles the \"keep track of which volume\nis attached for what reason\" calls to storpool.spopenstack. However,\nwe need to explicitly specify the access mode in initialize_connection()\nnow, since our os-brick connector does not know how to divine it yet.\n(this will come later with the \"attach snapshot\" functionality)\n\nIncidentally, this fixes the \"create an encrypted volume\" flow, since\nour _attach_volume() implementation never actually returned a connector\nin the attachment info...\n\nChange-Id: I5da7ae04b87b4fd52a682a6545060e852174f6c8\nCloses-Bug: #1939241\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/280335b901045b3fa94b3517db363b0e88290123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/280335b901045b3fa94b3517db363b0e88290123"}]},"branch":"refs/heads/stable/2024.1"},"e3f76c285c5bc2565c747c5d522c610b4841130b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-07-02 14:20:21.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/58/953958/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/953958/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/953958/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/953958/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/953958/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/953958/2"}}},"commit":{"parents":[{"commit":"0b50390766d19f15aacf82a91c36ce7c830dfc71","subject":"Pin flake8-import-order\u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b50390766d19f15aacf82a91c36ce7c830dfc71"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2021-04-20 14:46:41.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-07-02 14:20:06.000000000","tz":180},"subject":"StorPool: drop _attach_volume() and _detach_volume()","message":"StorPool: drop _attach_volume() and _detach_volume()\n\nOur os-brick connector already handles the \"keep track of which volume\nis attached for what reason\" calls to storpool.spopenstack. However,\nwe need to explicitly specify the access mode in initialize_connection()\nnow, since our os-brick connector does not know how to divine it yet.\n(this will come later with the \"attach snapshot\" functionality)\n\nIncidentally, this fixes the \"create an encrypted volume\" flow, since\nour _attach_volume() implementation never actually returned a connector\nin the attachment info...\n\nChange-Id: I5da7ae04b87b4fd52a682a6545060e852174f6c8\nCloses-Bug: #1939241\n(cherry picked from commit 8a75d036f65a8da335170910f1820308dbfabedb)\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3f76c285c5bc2565c747c5d522c610b4841130b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3f76c285c5bc2565c747c5d522c610b4841130b"}]},"branch":"refs/heads/stable/2024.1"},"27f168a9889b368a9fab649a705f71e4e4e02282":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-07-04 07:22:26.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/58/953958/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/953958/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/953958/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/953958/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/953958/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/953958/3"}}},"commit":{"parents":[{"commit":"0b50390766d19f15aacf82a91c36ce7c830dfc71","subject":"Pin flake8-import-order\u003c0.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b50390766d19f15aacf82a91c36ce7c830dfc71"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2021-04-20 14:46:41.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-07-04 07:22:26.000000000","tz":0},"subject":"StorPool: drop _attach_volume() and _detach_volume()","message":"StorPool: drop _attach_volume() and _detach_volume()\n\nOur os-brick connector already handles the \"keep track of which volume\nis attached for what reason\" calls to storpool.spopenstack. However,\nwe need to explicitly specify the access mode in initialize_connection()\nnow, since our os-brick connector does not know how to divine it yet.\n(this will come later with the \"attach snapshot\" functionality)\n\nIncidentally, this fixes the \"create an encrypted volume\" flow, since\nour _attach_volume() implementation never actually returned a connector\nin the attachment info...\n\nChange-Id: I5da7ae04b87b4fd52a682a6545060e852174f6c8\nCloses-Bug: #1939241\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n(cherry picked from commit 8a75d036f65a8da335170910f1820308dbfabedb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27f168a9889b368a9fab649a705f71e4e4e02282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27f168a9889b368a9fab649a705f71e4e4e02282"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
