)]}'
{"id":"openstack%2Fcinder~953966","triplet_id":"openstack%2Fcinder~master~I7dee3752f4483dd5a952ef2b944ab640ae810fa3","project":"openstack/cinder","branch":"master","topic":"bug_2074377_qb","attention_set":{},"removed_from_attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-07-17 15:31:33.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"13915":{"account":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"last_update":"2026-07-29 15:33:36.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-07-29 15:33:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7dee3752f4483dd5a952ef2b944ab640ae810fa3","subject":"Quobyte: adapt to image_utils fixes","status":"MERGED","created":"2025-07-02 15:47:06.000000000","updated":"2026-07-29 15:35:38.000000000","submitted":"2026-07-29 15:33:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953966-bug_2074377_qb","meta_rev_id":"1ccdfaf50dd4f2a73eb5526058ade54ad70da0c8","_number":953966,"virtual_id_number":953966,"owner":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-29 15:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"all":[{"value":2,"date":"2026-07-17 15:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":2,"date":"2026-07-29 13:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":1,"date":"2026-07-29 13:57:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}],"REVIEWER":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-02 15:47:56.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-02 16:41:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-07-02 17:28:22.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-02 18:57:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-02 19:02:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-07-02 19:33:13.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-07-02 20:59:06.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-04 22:47:22.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-11 06:14:12.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-12 11:33:42.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-09-16 08:36:41.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-09-16 12:08:39.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-01-13 16:08:42.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-13 16:58:05.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-13 17:29:42.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-14 01:48:24.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-01-26 10:11:37.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-03-10 15:33:00.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-03-10 15:40:19.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-03-10 15:47:36.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-03-10 20:03:57.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-03-17 11:40:21.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-24 11:38:28.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-06-04 23:56:20.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-06-08 15:53:21.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-30 15:45:30.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-29 13:57:14.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"}],"messages":[{"id":"1c7e7e47dbb250d2b9bd0483dbf5ce2cb98f44c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-07-02 15:47:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6a71b6b85ffc1b7ae1c88fdcfa35290a9d94f57","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-07-02 15:47:49.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"c948a5f847b99a49a3a903aabc3740fb8201b171","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-02 15:47:56.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"526c795b6922d6541a39721fb51703727e93742a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-02 16:41:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1132/.","accounts_in_message":[],"_revision_number":1},{"id":"9a843271604c6cd655f54aee133883049a2d50b5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-02 16:45:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1261/.","accounts_in_message":[],"_revision_number":1},{"id":"24acfab365b6428f5b759438fabd05a7852a664f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-02 17:28:22.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b48cb83284972c6ccedb819edec78ae9b926944f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 18:57:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98b372be68694d58868073bcf0f9083e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8918acb17c46a292ada5343bc88c85 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/553d247dec0d4672912d9884b8d23398 : SUCCESS in 12m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14b3183e4f9a43429fa3ec5752531177 : SUCCESS in 11m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/789205a45ac64b00b00c08c7780842d2 : SUCCESS in 15m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3f44937b76420ab1ab7799b282ec77 : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/346755b1ec6b43f0a45578fda78ad9a4 : SUCCESS in 1h 00m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d3a968c77d314161865a3b097e55acc8 : SUCCESS in 1h 02m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d61f1ebb4154ae094342a2d6ed76f6d : SUCCESS in 1h 47m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a3e181e4e7f451a892ae76ddd3cf5c5 : SUCCESS in 57m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dc70168cdd07426ea064f4b7f1891620 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/59dea5d9394446debc6248092b4a1d26 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/37edf9eb390946209f7b0147722ef05d : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7eaf9dbe2fb346eab3d6e135bedad36b : SUCCESS in 5m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c18152e8e0654ed5a55aa9a286930e11 : SUCCESS in 8m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bf6dcd03309e45bb9f22b2373c7e273b : FAILURE in 36m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/094c4f9ed1d642438ec701b202309572 : SUCCESS in 3m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f9e74788a4cb45609b85f4a3879c39df : SUCCESS in 1h 28m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fcf74fb885ec4772b1f8f3d82e9351bb : FAILURE in 3h 01m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8b08d8f602e0448ea1bc33ab1c8d3671 : SUCCESS in 1h 37m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c45a836153cb4b62acddedde38be367a : FAILURE in 19m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/adf99256b65748bc8dbdd47282608617 : SUCCESS in 16m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0b3c9b851a1e4d9cbd48b79f96e6958c : SUCCESS in 1h 06m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/33598d5eb0304f259e921d7051305462 : SUCCESS in 1h 24m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3349790af9be49e8a3389a0bc18cb527 : SUCCESS in 21m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4d83b09803a7420a9f7d364ce5437a27 : SUCCESS in 1h 21m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3a782fabc11c4987b8bd6161f262a1b2 : FAILURE in 19m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad4005ff9ee4415b83bede7a1676c208 : SUCCESS in 1h 45m 13s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfe9cabaff0b421081ce283c715f6c2c : SUCCESS in 1h 29m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b9df5fe7f774459baee63dba2e4c202 : SUCCESS in 29m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a6a70011023ded547be63cd7579456f3f3791f7e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-02 19:02:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_07_2025_18_57_47\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"85377c41a03cada62afec3e8b83c2b6a31f6bfe2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-02 19:33:13.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d4cb66648cdf4d4b82a33e49a85aeb17\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/43e2ab47e36b47e38fc5c5bfbcb7bc27 : NODE_FAILURE Node request 200-0000070169 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c7c9e250c0f74ee4a20a233a61627cab : NODE_FAILURE Node request 200-0000070170 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7f003ca16af34affa88d31223767395a : RETRY_LIMIT Host unreachable in 54m 24s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4901a4e94fa54642bf529c2ed9214b59 : RETRY_LIMIT in 25s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ca7860905394c7ca10665565005bc88 : RETRY_LIMIT Host unreachable in 1h 01m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6b626186028bdaebec0a9acc2526c0a1aa1f9dfa","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-02 20:59:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1e8ac82e3597e98543c8582413eaff7c7daceb9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 21:15:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2113c607404002912807e235a80683\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ff07097fdbc047a3a573e91be8435460 : SUCCESS in 23m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/186016d4f3bd4d4f900e1aab4ec19840 : SUCCESS in 29m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bc1d7226a38c4a5692a98a2fc91701a5 : SUCCESS in 35m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7f54435ed013d7ad8f20ae25e61d3712528dc8b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-03 00:17:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"397d998d534732f918ee55c032525ae514244fea","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-07-03 07:37:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd4542bd6a3b0891e2aeb3299e993b470b053fa1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-03 09:26:14.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0514042425b4704b27204b0077747fd0862db009","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-07-03 12:43:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"562b7b1e077abbf5a7642c7ed5b65efcaf8be1c3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-03 14:30:35.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"db52c0a62173c2b1d1b51782b48ccade46406f00","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-07-04 14:22:37.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1ec6f53880000dcdb2c3cd2403d1d7616a9fe0f8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-07-04 14:23:02.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"44b61a49c096612d47256556b2357ae078f55875","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-04 14:23:24.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a551cd9bcdc7240ea6c371ee40c325b317fab222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-07-04 14:39:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ec18a75ef55a417d902352dd645fcab3a46ddbd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-04 14:42:04.000000000","message":"Patch Set 3:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d852bb31e2d4cbb6296fd5b78b842f4df558c0a5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-04 15:06:53.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/934d6b053155450495981ba66d07892f\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd33fb46f8dc44c59916e73f0a8554b1 : NODE_FAILURE Node request 200-0000070935 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5cc2b6a7b5444a8997118bdc159c2dfd : NODE_FAILURE Node request 200-0000070936 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3550e58d42d7454ba129b5e33e681415 : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/769051ab97e543ffb6e1f33f5ee7864d : NODE_FAILURE Node request 200-0000070938 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/289026598fcc45baa9d28f1382c2df06 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":3},{"id":"6b67b8379186938d14e3e6fd615c04507710395e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 16:01:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f35dbd20b0c47bbab8a808c15afb085\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2830bf2ef51e4002a72e64d01670cdab : SUCCESS in 23m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3a23bc49a6ce419b9e6a831f71e92cc3 : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/809b8135e2fe466fa7e35fa0428f4b2d : SUCCESS in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8868d40b949c0be4763881fb05f41572bd90c022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 17:14:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28a9231ca102419f910e7815e9f04797\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/164d9524507144f5b916830101092ea8 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6043826f69604b4a9bee55ad25bd72ae : SUCCESS in 10m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1fca7649eb646a9a052a9274bbef8fb : SUCCESS in 9m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24881b4c82054a69b7391e3a315dc405 : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3411a3bb364112b5425359dac94eae : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/37b32711a75845548f0caf049018e0e1 : SUCCESS in 51m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf9a9f5db6b046bf813267e8da4e714f : SUCCESS in 52m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0bd59c659d9b4c7d9837db1c0a652d31 : SUCCESS in 1h 30m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1eb5a85928a4d6bbbe0680fbc0f90bb : SUCCESS in 54m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/911320de56364fe497f5544f7f40d9a3 : TIMED_OUT in 40m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b4ba989b3793403591bbed8be0c7e3ca : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/39230bffcf044651b49ae0b9b112905e : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/acdd09b2f0ae42dfa779f2d2387bd04a : SUCCESS in 5m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e05202aaa8994ffc944f18a14d4484f7 : SUCCESS in 7m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6cb061444d294704ba831e53c1c62068 : FAILURE in 32m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e59bac98381840cf85e3166b6f297b06 : SUCCESS in 4m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/14e3694ea67846aeb00a06365d9f9fe9 : SUCCESS in 1h 41m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8f1b2e7d5f9c46548a3c6d541243c3c1 : SUCCESS in 2h 23m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b0e7205f9b5c4e6a9adedf853ee8eb75 : FAILURE in 1h 37m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cd4570c387d341e381ca1e13e9f9a2d4 : FAILURE in 19m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f9c882be1b3b4e5fb3a8908f584c6c23 : SUCCESS in 28m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f5d300e3c3444cdb89ef6126d8fc201f : SUCCESS in 1h 04m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fcf27fef7ad94f1086f258e5fd6ea393 : SUCCESS in 1h 32m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7cac32b7bfb247b8ad90447874eda94e : SUCCESS in 23m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9656440633f045d1b6bef43d477ddeab : SUCCESS in 1h 28m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/aefe836be3cb40b982e1dc9b1be2a89f : FAILURE in 11m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/338d3ba0a27b4cab85d8ce7423b02eac : SUCCESS in 1h 04m 12s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d16f535c44034edd98e2cc61899158a2 : SUCCESS in 1h 28m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41af04dedc8645b38ac784cfd85cc4d0 : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":3},{"id":"98240fa4337e8684925cc06115c844ec30c32f89","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-04 21:45:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1296/.","accounts_in_message":[],"_revision_number":3},{"id":"f44bdde75b2b220de4e726264eaf187f36bd208e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-04 22:02:38.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1166/.","accounts_in_message":[],"_revision_number":3},{"id":"1f36581549a767365200536345c772006604ab15","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-04 22:47:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/953966 : FAILURE in 35m 32s","accounts_in_message":[],"_revision_number":3},{"id":"5f2f4a70abe29125ef0ff40985f3b85a228be375","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-05 01:14:47.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"de9cbed43fe762213955ce4c825c866125912377","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-05 04:50:45.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"f434ae6bc5fa576bf786983836235981d77fe4f4","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-07-07 08:07:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7dc0fcef28ef3029c8516ef0de1aff9ddbab2d88","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-07 08:08:20.000000000","message":"Patch Set 3:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":3},{"id":"df7a35b8fe71c8b419c91743b7f06e8869b114ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 08:40:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4028f406c8e4e87b1cf5397c17a6a58\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/16e2da85c2ff45d1b6eafb14deab44c2 : SUCCESS in 22m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9628510c3bb34f34a8732877a843e7c3 : SUCCESS in 25m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2db6db21030a453c9c6894ca53e362c8 : SUCCESS in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4a562021e1f3631de7aff44f74151a7e02f8b5f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 11:14:57.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc4c017fbd254b65a57b12f64a569b91\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32e20acbd8db414a8bb0ff589e4ec3b7 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c63ff72a9fe47b5bc31d0dd16e50994 : SUCCESS in 8m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f645c30eff943f584a84b2eb6c7ff81 : SUCCESS in 11m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be1b69c72a0c4d5c996e8c47c6f172ff : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a0aae7b0604ead88242f96d1a81cd3 : SUCCESS in 14m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/7633de5364724c1fb6e748978dd5be10 : SUCCESS in 59m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4fcd05cf9ef74356b33f5d41742bf136 : TIMED_OUT in 3h 02m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d257397730a4aeb9a92cd631273b358 : SUCCESS in 1h 08m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1d3828cb5c441b8969d4d329d53f859 : SUCCESS in 1h 01m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/36aabca348a042fca2f81c4382047077 : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/51c2191ff4f54b39877b74043d742311 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d58a8c7b7a824cfcadb085e19873ea2b : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/22b062502f29441991fce1bd52ee8fca : SUCCESS in 5m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9c5637828dcc4e03addf03b8aadf4c36 : SUCCESS in 10m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/747e2e1a0f264ffca5903fce54adb94a : FAILURE in 35m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/26884197338248e68a35da65e594cae8 : SUCCESS in 3m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d142747ce5ba41a19e9265981ea6e55e : SUCCESS in 1h 56m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9d2e3ca14f2a40d99c5af32ed32d1f48 : SUCCESS in 1h 24m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/37476f21f4464f4ebb8e130cc67f44f7 : SUCCESS in 1h 37m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d4c173ba3aea48319e0bac41f03cb4bb : FAILURE in 20m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/66e7a707969746808f508d797669663e : SUCCESS in 36m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d5958d6103a6435f844e6384c1001f8e : SUCCESS in 57m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3c3f20bfafb148bd8fa7b0b66f2a6331 : SUCCESS in 1h 49m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f170e59d811d46d98f85dd367d0f07f2 : SUCCESS in 43m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f202e54262df4aafaf90ba118dc783f4 : SUCCESS in 1h 38m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0304dba9217a4ae2aef28ffa99ba6e24 : FAILURE in 20m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0dc521ed534471797bc75022b422162 : SUCCESS in 1h 51m 20s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a9a7b40bcc342bab6dadee9a849e037 : SUCCESS in 1h 27m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5258df34fb324a9d87e67b216d471ab8 : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":3},{"id":"d7361825ba620251310f68c23b52c485f0b84fc5","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-07-10 10:41:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22cd390bbd5c95702eeff78ac43f7b8a5696fa74","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-10 11:09:06.000000000","message":"Patch Set 3:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":3},{"id":"078f9b6f37ee755fce7ae6998f35a74465aad8f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:07:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51fbb16ab2db4f268816af2351fa07ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e529feca13c42d3a444d7d1ea1b9294 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/feae012398f843749f58f3516eec6383 : SUCCESS in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4108d3d8cb21410c96b9f553d7d7b0fc : SUCCESS in 6m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f83b5f3d1a443f0888d73b073c987e6 : SUCCESS in 15m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8cc700e9c1442eab87823ae07e570f : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/061505e14db44aa6b81f1137f38379fe : SUCCESS in 56m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4bf55d91093d4092a6863f4dffa3144c : SUCCESS in 53m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4bd73aa0066f42a0a106ac1589794bcd : SUCCESS in 1h 22m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b03042def886461a9a40f4d941aa7e67 : SUCCESS in 55m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c61c4ab1707d433f8de6a60c0801db9f : SUCCESS in 30m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c4599f2422ff4f1e9109c72324baef05 : SUCCESS in 6m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a840dfca34704f84a70a0c17aba6f0bd : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e3bcefba16164dd49426bc4bbec25ae8 : SUCCESS in 4m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/23b7a26ca8d94ad0b0130270690db08d : SUCCESS in 7m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7f28c65c1a6f4f37a6e893cf178d05d9 : FAILURE in 34m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dd9ae5b11da442eaa1b565d676677917 : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1428d14a662943b3bcaef6401ba76aff : SUCCESS in 1h 49m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/25329da668bb45e48fa9c32b92cf4ab5 : SUCCESS in 2h 34m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f0894befb4e8440baa63659c5b4c0c99 : SUCCESS in 1h 45m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cbbd0477eb9f4a78bab64b35bcba6096 : FAILURE in 24m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7bc1284221de497cb8e98ccac6f48088 : SUCCESS in 35m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2d7145d140fd4460bb9bd17c6698b642 : SUCCESS in 1h 15m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/32602f6771a646538f16f67939145e22 : SUCCESS in 1h 40m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bb324f192bf541eabdfc19dff1314dd3 : SUCCESS in 33m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aa8f0e58e48444a5b56a5c381709a107 : SUCCESS in 1h 25m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/458e9423bb714067b6da90e8b3199be4 : FAILURE in 23m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f85a6828df0446f998ba484b0e8c2f69 : SUCCESS in 1h 45m 08s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c8b63264a5e47efab4153d110d4f888 : SUCCESS in 1h 45m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de3399097adf46828f6b5ef9140036b0 : SUCCESS in 29m 21s","accounts_in_message":[],"_revision_number":3},{"id":"8b80c6f72426eabe26a9b71df67cf81792b371c8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-10 16:01:54.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"a7a3ed9ba0c3f0448cb35b9d3432ea3954d90a57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:20:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f96b19cd0f74bb5a8508415faa7a549\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b438485b12ad4412ad062295dae3d46d : FAILURE in 26m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/478eb799747d4b1bad1e86728bfd661c : SUCCESS in 17m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c7bc963aa5f74cf9bd76d0b8eaec2fd3 : SUCCESS in 23m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4ba255af397723747b89e3d9a4ba2b32c8989f31","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-10 18:14:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"a98babd9e974973169c4a170bf34a41bddda3abc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-11 06:14:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/66/953966/3/20250711085949/NEC-ISCSI : SUCCESS in 5h 50m 21s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/66/953966/3/20250711085949/NEC-FC : SUCCESS in 6h 14m 21s","accounts_in_message":[],"_revision_number":3},{"id":"ff88c8ccd03d19c85ec205d3e5f1d0e27e7a0766","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-12 11:33:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/66/953966/3/20250712144127/NEC-V-ISCSI : SUCCESS in 5h 33m 11s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/66/953966/3/20250712144127/NEC-V-FC : SUCCESS in 5h 52m 12s","accounts_in_message":[],"_revision_number":3},{"id":"b82efd4cfd59bf909e7dc048d3d00533b1ab044a","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-09-16 06:13:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"90c8808d916c865915a5cb1f411b39c2b4a678da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-09-16 07:35:52.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c605c7872139e6d1c1c04ab404041e00e51b3a3d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-16 07:41:30.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"c71f544ded2e79354aab567e4962ef73f133c10f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-16 08:32:09.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":3},{"id":"635085853f460359e4192a9e1ce5139666ac7816","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-16 08:36:41.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/593/","accounts_in_message":[],"_revision_number":4},{"id":"7462039c8443baa7ecbdeee1d4959232e1e12ab4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 08:51:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35c4960dcf8f4152a40cfbe21ea37472\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f6d015cb77b9443b8ca5f0958b66fff9 : SUCCESS in 28m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/064cc563c2594d43974f36affc55d3b6 : SUCCESS in 25m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ab8405bf308496c94a7d7fee8def64d : FAILURE in 37m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"95fa13624662f86e55367cc4c00504676b7dff03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-16 08:55:29.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/953966/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/953966/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"022c4090fe30831b926310138e45bb4a1ed05cdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 10:11:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7f68d198fd44792ad2ef99e960662e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3d4ffc9ae504c6b8f00e3caec3d01a9 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/698e73d4d5764747aee0c904869284fb : SUCCESS in 9m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69f95b7628714e48ada3b7c41bb367c2 : FAILURE in 10m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd5185dc63cb4b9f886575275cf4ca90 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cbd6d00e8d04f929340aa0f8e8286ac : SUCCESS in 14m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/88b99141dea247a7bfd888af0c4b57ae : SUCCESS in 52m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ba631248fa24d3ea4721baece2cd2be : SUCCESS in 54m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5fcea77db5d0411793700d80ac7d31c8 : SUCCESS in 1h 22m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/084c07d8900f4557a385ef50597ca32d : SUCCESS in 1h 01m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7a0f3aa913c94c549975c88e43ff40fb : TIMED_OUT in 41m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c14f887fce014a44bdd62db08690c48e : SUCCESS in 6m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a0f2161e750d4fa39528ab717984a291 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/217a46dff66b4286ab5c8b7c6f0afec5 : SUCCESS in 6m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/67b96e771a344d32bf9f8e659fc242bf : SUCCESS in 11m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2bf00925933a4181aa4863328ce411a6 : FAILURE in 33m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f8a41128256843d497b933c08c06cdb9 : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/66e48f25bb5b43238fb0222a090220ca : SUCCESS in 1h 36m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ec4c4829174c45a985ebca79fd7478b3 : FAILURE in 2h 19m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c39da3ee7eeb4dcd97258a78848280e1 : SUCCESS in 1h 39m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4632785c8f924bdca290180c04a6b2c8 : FAILURE in 18m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c216cd87c30e47aa9be8913a803deffb : SUCCESS in 28m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/30ec1c51efcd4ec0bed5d3478a94645e : SUCCESS in 1h 04m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3d288553fa094e2fa1a17abc66f2d78e : SUCCESS in 1h 28m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/418dfc8b65b1457da1187803c1a7ce24 : SUCCESS in 41m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/05cd6a126ce94b8a83baaf182f9863e4 : SUCCESS in 1h 37m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1c1fcf4cb72e4b9fbda1c61ec55bf7f9 : FAILURE in 12m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f7025b53125b49f5802ebc24defc5ce2 : SUCCESS in 1h 36m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00a5a34fb8c04046a7caf921346ec9f9 : SUCCESS in 46m 01s","accounts_in_message":[],"_revision_number":4},{"id":"ed19e92d4c51940ce69ff6b4180f56f6ce506ac0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-16 11:52:17.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/953966 : FAILURE in 36m 37s","accounts_in_message":[],"_revision_number":4},{"id":"47158394590866ca304f7be8fef759676aaf8f53","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-16 12:08:39.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/953966 : FAILURE in 36m 22s","accounts_in_message":[],"_revision_number":4},{"id":"7c6ce23fa8ef7310b506303a18590a046892bb9c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-16 12:17:31.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":4},{"id":"cc04fa92600842de759a5cb74fd6866313d0af90","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-16 12:17:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3dd612d8a68a491bace5ba5974c098a5 : SUCCESS in 44m 42s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a29d5fbde5564a77806ac731eec234cc : FAILURE in 50m 47s","accounts_in_message":[],"_revision_number":4},{"id":"3708e31991ef066166a69123e28190d46131bd11","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-09-16 12:35:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b0acaee747171446c5f7d22c93e2830028bebf48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 13:17:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/567c008be4f5491995b09dd9d6c64793\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/838d01fcaad340a68b02f0b173d3e8da : FAILURE in 27m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32aeb0f4d6fe4ed89e3ad3604e274f64 : SUCCESS in 29m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a0cb2fe25e934f69b9b9ce56788fbbf9 : FAILURE in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"112a1427db078a2359aec04ada7a1b6f1558184a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-16 13:58:56.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1938/.","accounts_in_message":[],"_revision_number":4},{"id":"fd8fb522534513632802b2bc16206907b3d268b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:20:25.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/058622a15f204a339632efcd6eaa3319\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bdead46cf04933a034d6c8276be052 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a732c34b3a194eb7b636614f2b5d70e7 : SUCCESS in 8m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7fe43dbcc14847b391680fdc3377bd09 : FAILURE in 12m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52d6d945c79045c2baae0d32a3346c6f : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980ab118da5f4c4ebcca2bc85cc4a011 : SUCCESS in 14m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/9833bcccaecf4e0fbfdd600fa1f79c94 : SUCCESS in 41m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b7d6dd21cb1f43dbbfa938b1b60abfa9 : FAILURE in 21m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2bbdb80af4b14bf5a08934e5f217053a : SUCCESS in 1h 17m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9cd1f0e7b34444e1bb7204dbbaf73a1d : SUCCESS in 51m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/efff7d08cf8d40698802b2356e0435f3 : SUCCESS in 19m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/86161b423d2f4568bdaae53cf424a201 : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/972f2b4a8f574e779a4984ff74a17472 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b85cbca0bfea491486bb14f38f5378e2 : SUCCESS in 5m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b08f8d776d2c4129a21c5fb071e2bb42 : SUCCESS in 7m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/049e0606630240ce89872e62ca6c7fa9 : FAILURE in 26m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e2f98cd6e7040d1804fb2197ac92a9b : SUCCESS in 7m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d446d491a9b84042b75e38cc2722432b : SUCCESS in 1h 36m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4eb94667afc54f019581cb09be9c5cc4 : FAILURE in 2h 37m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/adc3f13cc73b474aa846fdf0c600b062 : SUCCESS in 1h 41m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f2ecbbc25cf6452abaac8dd741a260e0 : FAILURE in 9m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a54be693aae7419d835d25b3cb979b67 : SUCCESS in 24m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/101f7d52607648ee8cc4d6361529152b : SUCCESS in 1h 13m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ae5290e66b124a8d8ee4d053208e4c28 : SUCCESS in 1h 43m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e8cecbce6dc048d4b7b144345434bc38 : SUCCESS in 37m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a1d3b81035c344e6abdfd7d9c0814676 : SUCCESS in 1h 15m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3f1450fe3de248688a87c440d292a1c0 : FAILURE in 11m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8835e79cf2d44325bd858f3a03a33835 : SUCCESS in 1h 45m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7be21ab877f14dd98516dbc373325660 : SUCCESS in 49m 18s","accounts_in_message":[],"_revision_number":4},{"id":"f0f0146efe732220bc02c7a4a1434fb60a21d762","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-16 15:20:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/658669b469814d39a1f96c3710d5a36a : SUCCESS in 44m 14s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cca59b198cfa4751aee344db81f35c07 : FAILURE in 1h 02m 29s","accounts_in_message":[],"_revision_number":4},{"id":"9fd488c6b5c467d1b6e24497d2ce7e9b1565b9c7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-16 16:13:51.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1804/.","accounts_in_message":[],"_revision_number":4},{"id":"92088cb15ee1e5366c1ec243f23183b2f5286415","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-16 19:24:51.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":4},{"id":"65e9dc840085afa9aa59d739ef1260a67662f9f5","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-09-17 13:38:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ed0375a0d621a60963cd796cf7d6e2eb2d95635e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 14:12:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5ca6895aa144407bc32ea2c5b8b283d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2b0705a54624a9890e95ab51621079f : SUCCESS in 27m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46093cafef514660a3c79b964cbee15a : SUCCESS in 28m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fe918d8d84314115ab71249c75d02821 : SUCCESS in 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"801e5ff870fdc44131db11dc4c414ea68294b76d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 16:09:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7659136642a0442aa10a7918e3874f1e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29f413166d584c3d9cff2a3fda8360a3 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2d275208add47f2b3988f20a0f9675e : SUCCESS in 11m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/934925b85ddf43c5a206e5dfd0fb1339 : SUCCESS in 10m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e55bf2fcfb154e889c9474ef883a575c : SUCCESS in 15m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cf47671792842cdb075d9a0b262ba4a : SUCCESS in 13m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/bccd1c6b2195455d95cec0c2c1b3eb1d : SUCCESS in 56m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e652e8fda434dcb8391f99816e7a033 : SUCCESS in 1h 05m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28155881c2f54ff68a46fb5570de9b45 : SUCCESS in 1h 38m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12ae7e25b41d4d26b719aad051d64600 : SUCCESS in 1h 02m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2ea54dd06fb3430a82307d8028d94c0a : SUCCESS in 21m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b60e4e3d41e34ec49482c75ff8742cb1 : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9453f04ec5424e058b7f379d51424121 : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ab618e220c664cbfaea7f0629301639c : SUCCESS in 7m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/649d27d38e1846569302d8a3119dd744 : SUCCESS in 8m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/25c8ff2788bb489e9ba4ced3ec15d1f6 : FAILURE in 37m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0a783a57cf0e45c0a016f1e0a0532904 : SUCCESS in 6m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/372be02047c344c9bf804b829174f4ff : SUCCESS in 1h 38m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/317495ed93194e8fb168051b8b3cb57b : FAILURE in 2h 15m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7b258ee1a3924290acebcca53c7b706e : SUCCESS in 1h 45m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bfb7cb0daeab4a6dabc861f9f46d0952 : FAILURE in 18m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2116cf38a99a49af8e12f3e904ded7a1 : SUCCESS in 31m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fd2a254193dd4d7ca2e924051051a27a : SUCCESS in 1h 01m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/49f25d1166214a71b7be157ef488f8ee : SUCCESS in 1h 29m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1916c2cd0ab548649e4bf486f3f31329 : SUCCESS in 35m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6258d297cbb64a54aa5636d1e5f69611 : POST_FAILURE in 2h 02m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1993b43ac79f4a2bac77cc8f960b5f86 : FAILURE in 11m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1bc8abca6da243b3acc2fdbe96c2f877 : SUCCESS in 1h 33m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8dbe051ce304ef8a7e93ddfd9d81b27 : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":4},{"id":"cd75a2373eeb0ae28ecebadf3c5b8b38a395f3e7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-17 17:47:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/527b4b7637574c009958d94114bcaef7 : FAILURE in 48m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4afdcade6175487e8036f6b9ea856dda : FAILURE in 1h 07m 17s","accounts_in_message":[],"_revision_number":4},{"id":"f885dd7781c2aa93836bc185d1a647d0c653f24c","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-10-31 15:16:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"891dc2bf5d3e52c8d2bab01e529fecb7cc56ff92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:14:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/130b669d2708435d92005d10a8c2861c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/16aeac43918445739453c712192fc9ec : SUCCESS in 19m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd25cecbf09b4d24b52143c1be6277d0 : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/13368c67572a49e8a27f66078d30fdb2 : FAILURE in 27m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4f266fe9c13eedde487f7a925efd1c84cb51e98d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 18:07:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad740197b52042358e68cc1f6696c614\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fb44de8943e4115af1ab527dde92210 : SUCCESS in 7m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7aab9bbab3347f9a9bf12c95525485c : SUCCESS in 11m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34b44f5464454545887154f55715e0c9 : SUCCESS in 18m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a05262dd87545279d6a038ffd266360 : SUCCESS in 14m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/34ab2f86bde04e2581d9bfdcac11da35 : SUCCESS in 59m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/543269dc8f1a4c41af0b70ca80990783 : SUCCESS in 34m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e816ea69e38e4d389030178552d16648 : SUCCESS in 1h 28m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b74ebefe4f94b9f82995a2917b374c3 : SUCCESS in 1h 01m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fa30e0d620d24f339e4b8ca113ab6f69 : SUCCESS in 17m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/925cdb86d0ea4210aca9a75958252000 : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0130f9d34a334c2c81be94cc5ccbd658 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/40e6c9b89d7e4b21a728ba32e10f8ff7 : SUCCESS in 4m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3550b1c7479642c6af15c357e263d675 : SUCCESS in 7m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/963257b7f6a34a7ca973cf69679177ff : FAILURE in 34m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/86ae9d5773cd4cbaabf3fa67e3c826b0 : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/81a9891220844d94b2934c8642c0aa37 : SUCCESS in 2h 05m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/46052dad21504c1a8d6bfdc1d85c7b3f : FAILURE in 2h 46m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a460865496934baf8bb3fc245b4b88ce : SUCCESS in 2h 06m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/edc20c940527492b9398a4af2e83e1ba : SUCCESS in 2h 08m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/430ecc25bf9f4ad7b57f00bef351272f : SUCCESS in 36m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4af92ff0af3e4b819b6c7d5715c9a615 : SUCCESS in 1h 04m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e23a5b7a5b6b48dd81c41ebaf79bafe9 : SUCCESS in 1h 32m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/03a3c4c9273a4376a36e4320c65c637d : SUCCESS in 35m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/89125ac592f644d894adaa42e9685bf7 : SUCCESS in 2h 07m 52s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3238a396e5424b8a99f062e170a579af : SUCCESS in 1h 50m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a6f4029e14a47f595343653f602a5bb : SUCCESS in 2h 03m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5d3c5dced9e4fa1b879a7657e378034 : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":4},{"id":"af539f97d6834a3ab12b6c10b7219283f19e5a60","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-31 18:14:34.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_10_2025_18_07_50\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"d519aab90f7ddc0d1aae157b42fec53c3f872d5f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-01 03:55:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/01631bf6f73b4dcb9bb271353a4bfc82 : SUCCESS in 46m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d6c5c8e2c3ca42549de2452f7b3ae310 : FAILURE in 1h 00m 00s","accounts_in_message":[],"_revision_number":4},{"id":"9c5026a41cdd1e3ed02483fc96948ea58621c9e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-01-13 13:43:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3c5c4b43403d653fb4cdb5073ec99db540f20df6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-13 14:44:52.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":5},{"id":"1be345376df4a5c298949e13ae5d0e63616ff685","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 14:45:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/908180cf6265403b91af85388028aede\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ccb8372808d46029a1c312f12f9209a : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ea2a4f241ceb4426b1f9e76784157cb9 : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1006eee29f124f7e9d78fd116ef64b82 : SUCCESS in 19m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2877f7c2e43f2815a2a74c10179f915737fac966","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-13 16:08:42.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/804865840e1143c5914920d0bfe9f205 : FAILURE in 26m 18s","accounts_in_message":[],"_revision_number":5},{"id":"4d91f45f982a14cdbd2be7d61b5e6220fa8daf12","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-01-13 16:38:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"53eb816191bcdd22c3ca4d9f6710da8d051e039f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 16:44:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b62b02166f894cd88a1bdf3474117aab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba55b2b58677440ca69ad9664e9df3f5 : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/943719017d7e4cb5bdcbde12368d62f8 : SUCCESS in 12m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fe38756398347efa60c887e5a0661a4 : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19cad0536db545e48c42cee8df2ba04b : SUCCESS in 13m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/f870cf11322b46899f3d49051809f027 : SUCCESS in 58m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/58f8cdb487ac4f839af1264f0c4540d5 : SUCCESS in 59m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b284fa55b1dd4fd6b34b3b06ec40ef0f : SUCCESS in 1h 31m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f47500c2b7d344f38cab8f2b24e4c5c4 : SUCCESS in 1h 01m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/247a57a01fa94d54bfdeb9f3bb879d9b : SUCCESS in 17m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ed9904939e47409b9982fca5328d559f : SUCCESS in 6m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/decc88b25fc3489998221365d4c1b0d3 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/22e9467c2cca43b3a477e8af044a6bcf : SUCCESS in 7m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a79b106f11004388b3a413f621109187 : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4316514ee6484628ae87067e79737a77 : SUCCESS in 36m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/154c73638d1f47cca5496a466c1d7948 : SUCCESS in 7m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/46a5b3d74ae644d6bfc535226dae4b7e : SUCCESS in 1h 51m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/251b44552fce421c864a1fc8ff345133 : FAILURE in 2h 47m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3e91e69a4b6a4d5482e0fe83864be3e7 : SUCCESS in 1h 55m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6376e5b1d6664bbba2a2b8d25256dc1b : SUCCESS in 2h 14m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6824aedbfaf3473c80b22f1652d8c03e : SUCCESS in 35m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/67380c28c2de43bab5cfb0f60f5266c1 : SUCCESS in 1h 08m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8b4b75715be94d52a19d53a57bc3217c : SUCCESS in 1h 26m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/51eff2d7d93343cc8778b225ed2fad03 : SUCCESS in 50m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a1beac00790a4e089b9c0be4fdecad43 : SUCCESS in 2h 05m 04s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0a176c54e5f4471faf2192815a17075e : SUCCESS in 2h 32m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d40b18d7ca74c1998773f4a841cff5a : SUCCESS in 1h 53m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9cd7a7483d2496ab14a11cbcff546c6 : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":5},{"id":"4359f289de6f2fbd4aaa70f547fa9763d8060e40","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-13 16:51:35.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_01_2026_16_44_59\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"c85e1b0c97bd018c16bc9768b9afe04c190ae08b","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-13 16:58:05.000000000","message":"Patch Set 5:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_197/index.html : SUCCESS in 76m","accounts_in_message":[],"_revision_number":5},{"id":"67a91cfc61aa23e5707a013da8da2ad43fe17f20","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-13 17:29:42.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/ff9713b4a9584c2b82111c1a9765335a : FAILURE in 34m 12s","accounts_in_message":[],"_revision_number":5},{"id":"197a653a588532ed7bf0af42d4dd4db882903b62","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-13 17:37:04.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":5},{"id":"f252d3118a7026a732053c1913cd4053ced8039c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-13 17:49:55.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/953966 : FAILURE in 35m 24s","accounts_in_message":[],"_revision_number":5},{"id":"6c36ccabfadba35d9293b68e2db303836ea0fbc3","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-14 01:48:24.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-66-953966-5-fc : FAILURE in 1h 36m 25s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-66-953966-5-iscsi : SUCCESS in 1h 15m 31s","accounts_in_message":[],"_revision_number":5},{"id":"8f71e19448c2a1cff97fd9711eb72127fa72a906","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-01-14 13:24:13.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f21d5fe0fd12c5357a2404e087ee4d55088897ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 14:01:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ead6fe641ab4386b50ee7302733f047\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d0bb8a2d3384b02bc43eb3361252367 : SUCCESS in 13m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/721582619d264f7fa0e6fdc4972333eb : SUCCESS in 14m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cedc5ca9b10e403f8c3b5daa5184d0dd : SUCCESS in 20m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"18a25de0138e8d7b32e971d6123bd0c1c63f5724","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-14 14:53:17.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":6},{"id":"b11bc3a9d5471b8dd3256068a41830dd045b9fd6","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-14 15:16:46.000000000","message":"Patch Set 6:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_209/index.html : SUCCESS in 74m","accounts_in_message":[],"_revision_number":6},{"id":"d9be1805492b05581748f8d87b639e299c9def15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 15:31:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa902c9046734f84854ba41583f5754e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8cf85ccbf0743ffb5e60fff85e027c9 : SUCCESS in 8m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75f47ddc1a1d4a9996dea0d8793f4d5e : SUCCESS in 11m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47fefbeecec0428cb461db82f4f333f4 : SUCCESS in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d0020c04024af689349dfcb814d9d1 : SUCCESS in 14m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/319b762522e54963b372d822dd451fe4 : SUCCESS in 1h 04m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/02ce4180fdaa4baba9382a132c151d78 : SUCCESS in 1h 02m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a6e39c314954241946474879443b572 : SUCCESS in 1h 51m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4377c6e0f3324e5f81755e729c16492e : SUCCESS in 45m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bc662e62d9354f9ab91d63b5fae6d4a7 : SUCCESS in 25m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/96f622a2d3f347489ea98800fdd1b313 : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9a84e7985f284b75bfe6e36ea20f3b0e : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/20c4737432ac45bcb3c115d12d2ae142 : SUCCESS in 5m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6488d9448d3e4fe3ae8227eaadf3013e : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a8dcf67449f944bd8dab40d044d27946 : FAILURE in 26m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3a1301f9122446cf96057300f3a39736 : SUCCESS in 4m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b99d0966b6074c8b9874c62dc3691a0a : SUCCESS in 1h 51m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aebf382ac99e4d46aacf0a43f6d1f729 : FAILURE in 1h 41m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6a7326ebff54d08a36adf6abf7aeff2 : FAILURE in 1h 50m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/25fcb244b43748e5855f11993c0684e3 : SUCCESS in 1h 53m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/95888fa9ce1d40b280aa59b70a02ef5d : SUCCESS in 19m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e65b41db38424addac11ad16043bcbeb : SUCCESS in 1h 08m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/51f038e7338f4827b85dcfb9e25449e4 : SUCCESS in 1h 30m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ed6fa629360a437e9850b7178ccb788d : SUCCESS in 27m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f57ec69887924bf4be37a4c9f5e03fee : SUCCESS in 1h 16m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2de8fd1807ad48409f2eec736b8078be : SUCCESS in 1h 59m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8bd24e3d96954634aea11ab8f1375aee : SUCCESS in 1h 02m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c806f16681b34263a71736201564a42d : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":6},{"id":"2290f6f4a5c637de497c9f999bab78046e5efc37","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-14 15:36:42.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/4a4d9628d8fb4126b3085408cb651ff5 : FAILURE in 26m 21s","accounts_in_message":[],"_revision_number":6},{"id":"2c3f94516786bd6d6c228a460d07f25c241410b0","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-14 16:03:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a66fe92e74bf4e9a9d72597033c865af : FAILURE in 32m 14s","accounts_in_message":[],"_revision_number":6},{"id":"899efd3eeaa66a8333e2912ac8977c258f4fd981","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-14 17:33:29.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/953966 : FAILURE in 36m 14s","accounts_in_message":[],"_revision_number":6},{"id":"e8b42802251f03d5df89d46833047aa624522385","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-15 13:02:13.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a0d40c343cf542e181261481e1f0992e : FAILURE in 1h 10m 17s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5f1f739590b44181ba9b2442d040e72d : RETRY_LIMIT in 14m 53s","accounts_in_message":[],"_revision_number":6},{"id":"2372a663b29a1f47d1ba39d6423560031c22f88a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-01-26 10:11:19.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"da3aa9dee6aac125dbc38f1ac7d282d6ae0edcc8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-26 10:11:37.000000000","message":"Patch Set 7:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/667/","accounts_in_message":[],"_revision_number":7},{"id":"fbd9357664dc7d7abe44f516e05b698328a75db2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-26 10:11:38.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/953966/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/953966/7/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"97209d96f7c20b3a843f0c171de93e746de7ddf7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-26 11:51:12.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":7},{"id":"fde8389270b4ebe265aa8d72c9413e960f4c918a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-26 12:06:35.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8b08b49b6f314786a061c4f9350d11a2 : RETRY_LIMIT in 8m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f3e6fd55045a41cbb0a5a27e9b3cfdcf : RETRY_LIMIT in 8m 59s","accounts_in_message":[],"_revision_number":7},{"id":"7144dd8f0b585725f4561853452fffe5454c2109","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-26 12:07:27.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/512f806fa249420bb2cd4b31e083b5d8 : FAILURE in 32m 45s","accounts_in_message":[],"_revision_number":7},{"id":"5a31e8843db07ef6540faac25d3f5a483ad3952f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-26 12:14:01.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/953966 : FAILURE in 36m 53s","accounts_in_message":[],"_revision_number":7},{"id":"92ef8b1ff852bbf9a5842da5d3b9ac27da69253e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 12:18:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d613456146ae44ed8984b1d0b1bdd4b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b83e68a6f4f04b95afda72751678d4a3 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58e22d5143994a388ccca5caacac5057 : SUCCESS in 10m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19b4ac5e0d5b4298a02131a6c0b59c1b : SUCCESS in 11m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e39c87e5f94e0b91d1932543344d7d : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/5d86b8c9b4ca4e2d8cf7ccabe695156c : SUCCESS in 31m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9172fded892e4416a8b50f6bd6300b09 : SUCCESS in 32m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b7485d89e55497cb93f40187942bbd2 : SUCCESS in 1h 29m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/315a98b3a92c4ec1b6b7e8a84c286bfa : SUCCESS in 34m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c469ff2cd7054d16a71d0546b615b560 : TIMED_OUT in 41m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3e729a327ca44468a5cc96a58732438a : SUCCESS in 3m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/730abc13517d4e5f87cdcca9e90a5941 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c3257d372045408d88b87864c8a8c3b7 : SUCCESS in 6m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/849bf2d81f544da2a56d8883cadddd89 : SUCCESS in 7m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5b02f405ff6f472b90bfc78cf17c66a9 : FAILURE in 31m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a1def5c4c0a4fee836c06745f643f29 : SUCCESS in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/86ddacf0356c4c65b2214fbcad171866 : SUCCESS in 1h 01m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f30a37fbd8bb4c568bb5a2222471c1a7 : FAILURE in 1h 30m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/148f5d72d2c3438187e4ced97b6c427e : SUCCESS in 1h 53m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/382d862a70994966ac8f8bfe820a22b8 : SUCCESS in 1h 52m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3eaf9fe844224531a94fe7948ec51d77 : SUCCESS in 34m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0d7a804ede0a417cbf306e8dfcf5bbb3 : SUCCESS in 45m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/de97ff762a9d4e05873479b8cd70ba7b : SUCCESS in 1h 35m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3a82464bc40041399e50bd6bc2e4e89f : SUCCESS in 27m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/287d7ff6808d43a7b843bbf47220547e : SUCCESS in 1h 24m 30s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f4c87b0016d84bc092109f01ffc14dad : SUCCESS in 2h 00m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0102f9c52fc740c3a79915f4c0c0cd38 : SUCCESS in 1h 29m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84099c22da514ca79d4d27a4a5b6b224 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":7},{"id":"ad425915702c0e0c6b549f329ad9ca26bd57e223","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-26 12:25:28.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_01_2026_12_18_42\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"5a508bfbf1dee71948c40afc940bdef4cc397d6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 12:39:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ae89be1e9954e768e47b9d1c478befc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f285771e62c8452e835ea0d46ee6d8af : SUCCESS in 19m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/579195b4a5094945b016f386720e3b6c : SUCCESS in 21m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4091f18d1fc44cd09b0ae2e5c7f202a2 : RETRY_LIMIT in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"09387683ef7cb7dd2d7fc81581c98649e05f88b5","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-27 04:42:06.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-66-953966-7-fc : FAILURE in 13m 16s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-66-953966-7-iscsi : SUCCESS in 2h 32m 47s","accounts_in_message":[],"_revision_number":7},{"id":"de6fede13d48244b1dfb22e26b6bb29492fb7504","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-03-10 14:27:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"dce58871cb090116a351efb9f943bb7cdd1fac5f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-10 14:42:55.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/858aa73cc77d4d8da4f418aadfd0017c : RETRY_LIMIT in 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ead5ad947ada4d9d9b5dd9f25d94e29f : RETRY_LIMIT in 37s","accounts_in_message":[],"_revision_number":8},{"id":"aba5d1be61b2fcd38b87bac52112cb22e6ee0eea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 14:51:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bee2f4a499984abf885cb3c10d4e3ecb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa53d96b47914cb7a2dee33aa832c930 : SUCCESS in 11m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd65588f52e040cd846cb653d8c404b5 : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/475560d0e09a403cb2e0ede9be65b95d : SUCCESS in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e414a6140f9b2a8de7a372320baf33c409147e59","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-10 15:06:49.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":8},{"id":"30f261d11afd92c1ceb166e0e0253b73c95f3bf0","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-10 15:33:00.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver build/d7e57abefbae45eb9282c2e21a6abe98 : FAILURE in 33m 25s","accounts_in_message":[],"_revision_number":8},{"id":"e76fffdca524716b1c7cc59d569f19e5b53c8b80","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-10 15:40:19.000000000","message":"Patch Set 8:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/953966 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"91a90b4efc1933a7824dca693105787690e6a5c2","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-10 15:43:43.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/bb94b1d0763d42469c9551c135793c46 : FAILURE in 13m 18s","accounts_in_message":[],"_revision_number":8},{"id":"120b89ac893d4b2d2856a58e9e139ce9dacaa7a4","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-10 15:47:36.000000000","message":"Patch Set 8:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/953966 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"1f3291869ce8181f0502a4bdd21aff63dab840d3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-10 15:49:36.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/953966 : FAILURE in 38m 40s","accounts_in_message":[],"_revision_number":8},{"id":"7348bf47a6f9350ea5de20adf1ee5e77808be2fe","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-10 15:56:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/ad59f283c6b6485a98842ad93a2e4c66 : FAILURE in 45m 04s","accounts_in_message":[],"_revision_number":8},{"id":"e3e41ea877222c40a4f6ec3b80473ccd3cc98a1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 17:45:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d634b76b76a41fbb9180a9198bee738\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b62a6da4d74a63bbf5938f582e0fa5 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33782f9410ee4e5a97a39a2fe3128c99 : SUCCESS in 11m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50bbef4508814976820d040bbe1a4325 : SUCCESS in 11m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb1f78949f4a417abc923e995c939f85 : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a0ac0975d56430fa9c863455757de94 : SUCCESS in 13m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/f1df7a23e0be4f838cc7c709c0a3d4cd : SUCCESS in 1h 05m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9029d6629ca46558e9c6999a38f5220 : SUCCESS in 1h 01m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38f5863a6f6d493e8ded56c27725cc29 : SUCCESS in 1h 37m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cf687c25c764f1b990af62815ad96d7 : SUCCESS in 56m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7739adf272414e4ca8d8f43a9a8600aa : TIMED_OUT in 41m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d9d0826024bd471494be681d9e469d78 : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6d7d8ef3c9c047b994019b3be0fa990a : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/df8085f6b3d149fcaa96822da066f5f4 : SUCCESS in 6m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7d2aaf7e6de0455fa86a06320e748578 : SUCCESS in 6m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5844827cb434498681de0c35b7a54b22 : FAILURE in 43m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/58e6effd74bf44c49c9b159f363c69d7 : FAILURE in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ff21c8ab841d44f996412cd70a5dcf50 : SUCCESS in 2h 05m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/309f581da7c54b0a971a7d08f5151842 : SUCCESS in 2h 31m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b0ec794939744574b2daf4f4d5c575ae : SUCCESS in 1h 39m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7d33145ef2354b14b0ea45e21f703894 : SUCCESS in 2h 23m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/568d78ffd6fd4cea9c2df2b9a7453ca9 : SUCCESS in 38m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eb70bed9bfbe44978f42954e3730c805 : SUCCESS in 45m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/86cfb569da824b3ca2c404da4c49816b : SUCCESS in 1h 57m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/61e0f64c7fec4c44ae7563ea049d6bcc : SUCCESS in 47m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e4d03cc3e5e84bf6a6f4aabc85a8ab94 : SUCCESS in 2h 18m 30s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e1ceba0b39704a889b47e22b52085f2a : TIMED_OUT in 3h 09m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7702b6fc8b8d4514a500043aed121107 : SUCCESS in 1h 34m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3da34c574c14f2794458348d75447b0 : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":8},{"id":"2a8e96ea4dda1b5ebeae79ae61d808ceb254d2d4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-10 17:57:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3419/.","accounts_in_message":[],"_revision_number":8},{"id":"956b4f675d9447dd843be44d54c6cb9f7b752df9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-10 18:16:29.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3246/.","accounts_in_message":[],"_revision_number":8},{"id":"adb01a0fba12673517505395e761d31e30ff6e3a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-10 20:03:57.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-66-953966-8/FC/ : UNSTABLE in 5m 33s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-66-953966-8/iSCSI/ : FAILURE in 5h 35m 50s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"de4e38c7ca69a009964d029f54c0daf053854ff7","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-11 02:37:09.000000000","message":"Patch Set 8:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1045/index.html : SUCCESS in 88m","accounts_in_message":[],"_revision_number":8},{"id":"c364cacb78d41873a3ab2cd62f916cc9470fc54d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-03-11 16:22:30.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"85d2755426e9784914ae6071f0c422196f7016cd","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-11 16:37:58.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/7ddddd07f8df43f682acd7e6d005e1c0 : FAILURE in 11m 06s","accounts_in_message":[],"_revision_number":9},{"id":"c1c47a3be99a37d9d2966c0e8f112adf74b57808","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-11 16:41:37.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2da48170f26c4863867ee7dd4feeded1 : RETRY_LIMIT in 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/eeeba937f0fd42928157c59310a547f0 : RETRY_LIMIT in 32s","accounts_in_message":[],"_revision_number":9},{"id":"941c21fa8f4701f7e66edf57c8f6db30d04748ec","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-11 17:01:18.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":9},{"id":"3743cccaff7498b28ccfa3e3566434144c101425","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 17:44:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22c45fdf42c84565bc6433c5f8fca5e6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5993321477fa4ff6b377a534d1f88454 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb863346188c4ee1b6c9ae86d9897010 : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d8723be6f9804d37b60e6a0ac338fb8b : SUCCESS in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7800877220abd70cac97bd18c919b5ee61249e37","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-11 18:18:00.000000000","message":"Patch Set 9:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/953966 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"488ac353cb95beac2c217542c2f9d916378bb7bc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-11 18:24:45.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/953966 : FAILURE in 38m 47s","accounts_in_message":[],"_revision_number":9},{"id":"ad2f99ef0712a8ad9610edec3db18bd1fe6f569b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-11 18:27:06.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/19b5a4a816764f0581b1aa7473494297 : FAILURE in 38m 04s","accounts_in_message":[],"_revision_number":9},{"id":"aa7d49b069b6d990e763c0b8d52b612d17b5608c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 19:02:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5103743c894b4c999ce13e6b41ac71ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7199cef0f0d14195a1fd70538baff7ca : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99585672e3074a59aa338674798b7ded : SUCCESS in 11m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af6e2633f0c742d09e51657e3f63f0cc : SUCCESS in 12m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e13f85a2644b4823b19a84f32dab48f0 : SUCCESS in 13m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb6454fc5a7d4d269179e0c6cb1f904b : SUCCESS in 13m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/b210ac1aea15497392f2b4f7f524578d : SUCCESS in 53m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e9cb4a13c51490a8156b36be61087c4 : SUCCESS in 59m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3c5f328b900442e7af4d8a80fec2a3cc : SUCCESS in 1h 35m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38d71e6833274e16b59cfae401676d2d : FAILURE in 26m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6dd6ea637c7e4e998ef9a718e0796f68 : FAILURE in 26m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a95466f2e781499083aaa45f846fd1a0 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d1ec9737c4004b48bd2daea83dfcc1f5 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9e583cd62c745c89bd5579d4fb591e9 : SUCCESS in 6m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a842dd6b06854ab2b7b5415063d802f9 : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/98510714c05841adace2d02d962c1895 : FAILURE in 40m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f141136b832419eb07cb4c11dfb5fe4 : FAILURE in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a5b5b4ecf8a487f8faabe8ba811603e : SUCCESS in 2h 01m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f01f52389fc345dd99418574d198fddd : FAILURE in 2h 27m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0990a830a4794fc980f19ded503d7309 : SUCCESS in 1h 46m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bf16e8b4dba644019124785876848534 : SUCCESS in 1h 55m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4ad33ec8b2f34eda85296be3535b1bf5 : SUCCESS in 37m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8cf2f038eaa3496a8ca397117789bb94 : SUCCESS in 35m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/482ff3d1bc104bfcb9feb196e51109b9 : SUCCESS in 1h 24m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9903dbc5ab0b46f49d9e1847f8ad9e63 : SUCCESS in 48m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6fe650ae67304489af809b31ebdbdc5d : SUCCESS in 1h 22m 33s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2856bbe83b894d77a8bab53268b9a2ae : FAILURE in 2h 12m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/23c0699b37f94d7ea5597e74244426fe : SUCCESS in 1h 44m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b7b3808d8914aafa8f8a57a167c0cd1 : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":9},{"id":"58ce430f3686402966b82169b1ca8c0e368dcd31","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-11 19:41:51.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver build/39caa318fa26435ea83d45f6270d7c3f : FAILURE in 31m 51s","accounts_in_message":[],"_revision_number":9},{"id":"ed1c3be2eaf3cb363df0e4a7f9dbdadf246e16c5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 22:43:07.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3271/.","accounts_in_message":[],"_revision_number":9},{"id":"5f468fa785d532d63ed55e22f3d64d3db67aeeb6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 23:46:34.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3444/.","accounts_in_message":[],"_revision_number":9},{"id":"5a3bd69349b24f4d715dbdf5d1637a8eafb2c33e","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-12 07:40:41.000000000","message":"Patch Set 9:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1073/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":9},{"id":"91fcebdba8aff00bcd4218ef1447a07ee077e3c5","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-03-17 11:40:21.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"e101b1659959b82b91ecbb3e5aeb92e777d0dcaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-03-18 09:56:12.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"068b79ff12b6dcaab58309ac569bbab38fab73e9","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-18 10:04:03.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/daf15801a8ae44c68b04c323618d8cc1 : NODE_FAILURE Node request 200-0000654206 failed in 0s","accounts_in_message":[],"_revision_number":10},{"id":"9da1619c1bfe0e0feba257000fd61660eee284eb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-18 10:12:16.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d66d6920ed934d41963ddff134425d9f : RETRY_LIMIT in 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ba30f8171f74499b94a87361df3f6f58 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":10},{"id":"a9e862caca5238d5494d68b2f64998ae78557abb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-18 10:34:21.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":10},{"id":"3f765805f467acfd8b724eea69d2d507f402cf1c","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-18 10:43:07.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver build/fc03cff129104ce49f053c893af9add6 : FAILURE in 16m 27s","accounts_in_message":[],"_revision_number":10},{"id":"17cd6617cebe3482dc7c99b792dea5c7332046f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 10:56:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3cd657619164b828b5bbd28c7b8a0d7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81c195a5ba834a1cb02f91995e928c65 : SUCCESS in 16m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/636c10df79e64fa4ab0cb54ea88d3f93 : SUCCESS in 14m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/aea8ee59c76b4fba863932fc1c8c5422 : SUCCESS in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4023ef49873ab5f06d7da51404c55584f3bd1c70","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-18 11:24:08.000000000","message":"Patch Set 10:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/953966 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"fdab8fba49e28b9ae3a1b050429df9c15556ed42","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-18 12:01:56.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/953966 : FAILURE in 40m 01s","accounts_in_message":[],"_revision_number":10},{"id":"ad0a6d0ab8fe672885be62a3d01e366b198f33c1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-18 12:07:13.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3519/.","accounts_in_message":[],"_revision_number":10},{"id":"3992c54dcadf65bdcc717a3c2269361a8dad605b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-18 12:09:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3346/.","accounts_in_message":[],"_revision_number":10},{"id":"bc68c219cb392dc7eaad2c44ddd25d4b0d41c8a6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-18 12:18:06.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/ef6484c28a9649e9887cff7118237337 : FAILURE in 44m 55s","accounts_in_message":[],"_revision_number":10},{"id":"d3b905c09d9ec72239c96585f6d8a944a21b6732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 12:26:35.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d20bf030695849e4afbcf60d76a4836d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d186110359ef4806854e74da10efd95c : SUCCESS in 7m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5b3557d25f0431687fad200e65e6710 : SUCCESS in 12m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8aa590764a94793a7b42ff1a5ed3d07 : SUCCESS in 12m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cb553d6c958246f5b3f5cc1e8b502b84 : SUCCESS in 13m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/073d7f53c0ed45428b8266cb24944757 : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/f5514dc0496a4dff960a046d64df5cda : SUCCESS in 55m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ffcb876202d44d17b84ab944664a6a59 : SUCCESS in 45m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8e014540b7014b319c5f0f7c2f134275 : SUCCESS in 1h 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/462114925d2348a2b3597c717e79ed06 : SUCCESS in 1h 00m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b2fbb3b18666452f8f3abc4a166c3a31 : SUCCESS in 28m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2482767366a5460887b6a3bc98932d9f : SUCCESS in 8m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7968e17ef1064bcf959c27af1c2cd4b0 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/01734291ce0c4012b56d2ef686b348d1 : SUCCESS in 7m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4bb0f32410f041529242bf905eaee7d5 : SUCCESS in 5m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/07d5c802a4a94823a4b630fd5d250e16 : FAILURE in 19m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/87f419328fdf40ddba34f86e4ae7343c : FAILURE in 6m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/30a4940c7b68454599817311660f307a : SUCCESS in 1h 48m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/57e552d8cc2e4e798c30205b0f840ee0 : FAILURE in 2h 23m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8dcc048251594f5caca5826710b69623 : SUCCESS in 2h 01m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d4ada925f5de45bc82691154f69d89e4 : SUCCESS in 2h 03m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dfaa8e0293ba48fba93d23e25133a2ee : SUCCESS in 26m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1b488dcf71e74809b129cc69c3a6695b : SUCCESS in 41m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/36c16ab4f6de450fbcfbf52d7cb7227e : SUCCESS in 1h 26m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0c41e46471a84a18a81370b20a3d30be : SUCCESS in 27m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2763a72011ed4e6393105104a92c50a8 : SUCCESS in 2h 14m 58s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d679302d3b314b12bc7460f95c829b24 : SUCCESS in 1h 49m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a87ecd7d2141465c948a31b15c2371e4 : SUCCESS in 1h 27m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e2d9379f2f54d909a664c12f4bd40c3 : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":10},{"id":"23a4caa231231bd99d8fe6176a9a0f6caa48fe3f","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-18 13:14:00.000000000","message":"Patch Set 10:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1162/index.html : SUCCESS in 96m","accounts_in_message":[],"_revision_number":10},{"id":"46e3b2eb865be6a998f7523a613abf9e94d31920","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-03-30 15:53:33.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"f9036f39d54b1ec0807d491f3eae72dc263eee64","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-30 15:55:43.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/00b3506a1cce4cc88b77bdc4a72cb5f0 : NODE_FAILURE Node request 200-0000690806 failed in 0s","accounts_in_message":[],"_revision_number":11},{"id":"50d9d2a6a439037b2c30aafa8803ae39c06c1924","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-30 16:06:41.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ea67d7441e7c4657a55989267ee824ac : RETRY_LIMIT in 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/72ff884840ab45a2b206024883a0eac0 : RETRY_LIMIT in 39s","accounts_in_message":[],"_revision_number":11},{"id":"751a67196877a31a2324095a1e2ef0d56890cdab","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-30 16:33:16.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":11},{"id":"f50956424777f036321d19b725f5c5def4578eab","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-30 16:57:34.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver build/96e9295841bd4961b2039f0d3e2949ec : FAILURE in 33m 11s","accounts_in_message":[],"_revision_number":11},{"id":"a2918518bf565b3864af60be41bb3372eb2cbed4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-30 17:18:29.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/953966 : FAILURE in 38m 05s","accounts_in_message":[],"_revision_number":11},{"id":"a2a4703801daa180ad5b1d8f94f877495be217f2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-30 17:27:17.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/bff80c06a8c7476494a9c4935596cc48 : FAILURE in 40m 31s","accounts_in_message":[],"_revision_number":11},{"id":"6fd62720df20327f19d41133d37ffbbb58442080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 18:23:30.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c27a045780204bea91bb9c3557b8ad1d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e4a98c871564473bfcec6cbf62dc881 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd617f95a9bc4c0488c41250fa1da475 : FAILURE in 12m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f64bd3f6a9541e9b2b282c39718779f : FAILURE in 10m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6d560c62b8ea46438994a40a027fbad3 : FAILURE in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f8b151a71c45e3b751caced6771e41 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/06df6933c9b9472294de516c4087efb0 : SUCCESS in 34m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85b1fa2363694c4598b326dcfb18e092 : SUCCESS in 51m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b59fb29ec8c74b9380f73b59458001d3 : TIMED_OUT in 2h 07m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/439d2f1be6664d9484aaecb4712aa103 : SUCCESS in 35m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/15dc214605a64dc89e5330f05fae89c4 : FAILURE in 20m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4387416375584c38b74658b3a639aa95 : SUCCESS in 5m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8791ecffa67c467290374a32347e1e69 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b4ac139c62a42eca7b4cc89095500fc : SUCCESS in 6m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1b9ad0e0660d4758b935a8568cba523f : SUCCESS in 5m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/51f7c27f92c942afa60846eef7af2209 : FAILURE in 43m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f01af90d1d4e4bf2baba03fe19a975ad : SUCCESS in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b2153c2cb76748b5b12c49d27962df14 : FAILURE in 21m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b82c6701c6764e1c9090a083fddd5731 : FAILURE in 2h 28m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/038fd9f4eaa045e884027f867eb0280a : SUCCESS in 1h 48m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/530b742ba21c42278270283c07228474 : SUCCESS in 1h 50m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0e91f3400dcd441fb2d1e426326de472 : SUCCESS in 21m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1769d884d4074da98b13bc062bfedeb5 : SUCCESS in 1h 16m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a17f7e5bf1914478adbb6ceb9499ce3a : SUCCESS in 1h 29m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/49fa13eefc5646f9b96ce72ce75613a2 : SUCCESS in 48m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f4198b4e5e6d44ecae6b19876ca56a08 : SUCCESS in 2h 09m 19s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/1721c0a63f084a5cb74aaf28fba3f03b : SUCCESS in 2h 10m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75662aeaf4d5423c85da628ff98c96dc : SUCCESS in 1h 29m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce531db7647640b1990f80d84b911187 : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":11},{"id":"f7be530dc5ade63dba402947c7a75cebb8e9f218","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 18:45:11.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/344bd2494a374bb88e40a11d69504606\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59c6eece511d4cceb9385bdd4fc11380 : FAILURE in 12m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/daed4f7dd0a941d49f85ae0650d1a5b5 : FAILURE in 19m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2d2c24010bbc428e99d7332eec2ff754 : FAILURE in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"45f7fbd67008b59d6834a2b55c7f7ab29bd86c05","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-30 20:30:00.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3626/.","accounts_in_message":[],"_revision_number":11},{"id":"adbf013f427474543755b417f964a7b2fb1ef3fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-03-31 14:24:05.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"07746b03c82fd4f4653fed60fe664f19f4eabf4e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-31 14:36:56.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b6795bf07c78438096c70871f27719c3 : RETRY_LIMIT in 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/961825cc142f4363a88a63db9537da32 : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":12},{"id":"dfae153aa0975a2f0c01f296982d868e6dadd095","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-31 15:02:23.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":12},{"id":"7c838ca1942f5a4b4c5dcf275a47e923d23852ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 15:43:27.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db367a0d74684723a27907ba879ce614\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7295fe2185c4c8a98b67858c34efb12 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2c8fc59b2ff947bea3d27f0d7707820f : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e0c5fae7cb514ebea72e83ec3cd91564 : SUCCESS in 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"74f3f6cf43b3293f53e8c739ef7efd3d4b5767db","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-31 16:59:51.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-66-953966-12/FC/ : SUCCESS in 2h 35m 25s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-66-953966-12/iSCSI/ : SUCCESS in 1h 59m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"aed209e4c106616488459b3715dd36e0fc5273ed","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-31 17:58:11.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver build/10bd4941264244c996277af5f90e64e2 : FAILURE in 34m 51s","accounts_in_message":[],"_revision_number":12},{"id":"59500b08bcbd749d605d451a41974c4fc2ee0f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 18:04:43.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ef63c81dc7d4a2e84a478e2e966c721\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c97a95ec47f54f3fa9b5fc3131fd9ebb : SUCCESS in 6m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be4572fee9f84c08b186082d3509bedb : SUCCESS in 11m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45881f39847c44dab2365f6e93c58092 : SUCCESS in 10m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c176dc9df874361b77f8e29141b5452 : SUCCESS in 15m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06aa0ede2aa54fe9a3a8e8b1d6352379 : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/4f21d9f31a384fe785debb0465d1bb8b : SUCCESS in 44m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/da2ac7970b934e8d8b3e547348513f14 : SUCCESS in 38m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dc7cb6bff355464b963b657288a95c18 : SUCCESS in 1h 57m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e2728a861204a3d89b6302aca136fa2 : SUCCESS in 1h 05m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d590f7239b3410f89c9b26fd8a5576e : TIMED_OUT in 41m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2bcf97e8eddc4f58a0ba4e3c2a90bc38 : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d9ba34aa54c84886aff42c278808f37d : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e36243f12c6c4314ab169cdb6fd73a50 : SUCCESS in 6m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/03a8c762a5f943f6ba67b199e22e0a4f : SUCCESS in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3417d3f4b95449f9a9ff9ca267b26a6e : FAILURE in 46m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/712e0c7fca024b85b61b53768f1d7492 : SUCCESS in 6m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/86250ad5af184a89a170224aabf44d43 : SUCCESS in 1h 20m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9609d2b404454c329260a17c1a42dcd1 : TIMED_OUT in 3h 34m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/100b02d9d19b42af9f1f99e824e13fd1 : SUCCESS in 2h 06m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2327d99f609e40c59c9b2561c6129daf : SUCCESS in 2h 01m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f0ca534a9274431b9d461c4e0cd29ede : SUCCESS in 35m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/833e790e13264ad8b5c03d29b06bb27e : SUCCESS in 1h 11m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/50be8ca402bf4835ac2cde4abed5195c : SUCCESS in 1h 30m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/64af0ec5690b41be95bfb3774eb24fe2 : SUCCESS in 42m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b3e4747bf9df4a73a3755c3501184567 : SUCCESS in 2h 26m 03s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/378ca29287a545579911740077376eaa : SUCCESS in 2h 16m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f13c3762fb1b4a3e91259cb7e82ae294 : SUCCESS in 1h 03m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b452245642824842b83b401efa25a103 : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":12},{"id":"640031aef3307c3fae28ec975fde5e8d74d895bb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-03-31 18:14:04.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_03_2026_18_07_44\nLeave a comment with","accounts_in_message":[],"_revision_number":12},{"id":"774aa1808e01bd103a3d9cf9d439dc436c184e5f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-31 19:09:32.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/953966 : FAILURE in 38m 00s","accounts_in_message":[],"_revision_number":12},{"id":"67ea4887405297bfc2aba96ecf41df5edaad6d88","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-31 19:51:12.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/905ed580223d471cabdce4c0934b8259 : FAILURE in 39m 11s","accounts_in_message":[],"_revision_number":12},{"id":"e6ecf2cb93ab4a752a0e45e562d67d677d4923fb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-01 17:24:47.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3655/.","accounts_in_message":[],"_revision_number":12},{"id":"18e08b15152714b46045c4eb6ddb818d520bfc77","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-03 02:23:00.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3453/.","accounts_in_message":[],"_revision_number":11},{"id":"2a873c131abe157dc3ef50527af9a39c6d51f973","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-04 10:03:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3482/.","accounts_in_message":[],"_revision_number":12},{"id":"61665682160c37a58e6b3d9591b0ef3581d9aff6","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-24 11:38:28.000000000","message":"Patch Set 12:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/953966 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"99935810942fe2c4e30ccc24938c89bf91a5f144","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-06-04 23:56:20.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"886be8095c069ecaf764c41532ef93d1560f8fa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-06-08 13:27:15.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"5c359de9aa0187d9e3c6bf88b676a63d7f64eb91","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-08 13:27:27.000000000","message":"Patch Set 13:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/872/","accounts_in_message":[],"_revision_number":13},{"id":"456589b07241fbf159e52d71bd7f5d45572c29da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-08 13:27:31.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/953966/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/953966/13/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"2e00663a22d51394e4cb89245da1313adf75a72b","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-06-08 13:28:13.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"fd57a1e1a28bed5c41d9a4652c544c39933f5f79","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-08 14:26:31.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":13},{"id":"91f3bc8f75e47f4290fd748842c1bf7129ec7822","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-08 15:49:56.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/028f9efa98aa4d5080460dbeea4caefd : FAILURE in 24m 29s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/953966","accounts_in_message":[],"_revision_number":13},{"id":"d58068b8ecd9570550be5cc500a88c271382927f","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-08 15:53:21.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver build/c23e574af3204734aad4b6f63d0fe77c : FAILURE in 36m 01s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/953966","accounts_in_message":[],"_revision_number":13},{"id":"964d355a7d66844c5a03c725a8789eb87acdb6fc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-08 17:15:16.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/ac7c4916c747457ab58340579109e873 : FAILURE in 39m 54s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/953966\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/953966","accounts_in_message":[],"_revision_number":13},{"id":"7ceea2580ace96a1415af6c807abede9049b0fab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 17:24:32.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1046dbed38404910b88c9d671d87c1da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5267e5f402f4b7a94b781ce13f10b34 : SUCCESS in 8m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/240a8c653c1f4baf90f8183b10ffad03 : SUCCESS in 12m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/243d81aa26444a688b63da5a6acce006 : SUCCESS in 9m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/18d9ab9d115646c8ad802c2cad17c406 : SUCCESS in 16m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a9fba7683f46519c7f782c37ff8a5f : SUCCESS in 15m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/6bbd456e4e454ced8e312df72fbc1b52 : SUCCESS in 1h 08m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab42f816d9034d22bf0d830b147fdaff : SUCCESS in 39m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e3434bab2a7e479391370fe0c2877e94 : SUCCESS in 1h 42m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3681e26115f04899ac30e93a75d100b5 : SUCCESS in 58m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deabe964162b458683d2d0288fbabce8 : SUCCESS in 9m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e536229ba47446b585abcc670827161a : SUCCESS in 26m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7cdf0c7e586b4bd29f128511f0189283 : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8e57d9f3b25641eea0fba531eafc5239 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d4c41a2ed50045a79476eac6e166e1ac : SUCCESS in 6m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2bfaa6265f2f4a6da5bb66fd7bd62b75 : SUCCESS in 6m 13s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f9bd3fc72b2a4b9cbe8b15de4cc65cc3 : SUCCESS in 9m 56s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3d2197cec0f44a48ab881e83ffd34fac : SUCCESS in 40m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/53e3d66b4e3a4251a5d8706cd9ba1f2a : SUCCESS in 7m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1f636892858c4a68a3b4b72fe4b36316 : SUCCESS in 2h 15m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d63ac780d56944f5a8b848017da410aa : TIMED_OUT Host unreachable in 3h 45m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/160030b10f4c4db1bf179c9a606f7ce4 : SUCCESS in 2h 02m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4b9f6ff1f91d4b09b12647cf10fd88ff : SUCCESS in 2h 10m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/75afb8be4f564e65a55ef784af229a3f : SUCCESS in 23m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/15fbc1dfc9f44fe9b08af9615a7adeaf : SUCCESS in 1h 13m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cc5dcd4d811f4e219bb6eb7bdd0d87ce : SUCCESS in 2h 00m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d5dcfa749a4a460b9508c0f9407f45a5 : SUCCESS in 46m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a3d229698fd4fb29ae1ba5eb5dcbef8 : SUCCESS in 1h 50m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1ccaadf7dbd4f48a6fefd078a339750 : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":13},{"id":"e1670f7cb66c9f3234d55faa39952797e61f840f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 17:58:45.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82057b60469648c2b751079f50743502\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/56a392ca28b0467f9c398ed81dcd5695 : SUCCESS in 16m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1cd2fd5099c44a178d4c5191733525d7 : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7f37f463f30c4e19b3784c5acdd2725e : TIMED_OUT in 1h 01m 09s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"85510baa3553b3c92ec2f524e49b533a14fc4c74","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-08 17:58:45.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/08_06_2026_17_51_32\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"f94b7fe110e03240c2b9114514a8716fe204a5b7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-09 02:17:37.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/2143f332eec5450b8e6f025c3945ac7a : SUCCESS in 49m 05s\n- cinder-storpool-tempest-iscsi-multipath build/66f3c49c57d3415a91138ae6f93f2a80 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":13},{"id":"7a0f3e022659d02fbe80b95ef7c2e552d73dfe47","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-09 04:36:23.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4395/.","accounts_in_message":[],"_revision_number":13},{"id":"c99bcbd7e78bd4a82844a6abd1a357fd396ce88c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-06-30 07:43:14.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"44d60ec3d2f4f7814427a30cf0ad5d0f1114906e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 08:34:48.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d60da777f96244e58cbe94ea855f02f7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/81def220a49e474c80242b1e10439b0f : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f80ff76897d84aa794d73d3ee046027d : SUCCESS in 19m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a496742812de4fd38c52b9518c49d362 : SUCCESS in 21m 50s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"63efa71fac1f672cf23c55764b5eeaeab5036b96","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-30 08:46:41.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":14},{"id":"b750e8c06c8dab2fd22c34394dbee9bdabafab5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 11:29:27.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7be4dfff274d4be28544fd0d790d6102\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cedace07bdf4ab1aad27bfad2ee81f8 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21743cdcf5b64ec5bd6ae470546d4ab1 : SUCCESS in 10m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9832440d6876492395d4be9b0d4d2a87 : SUCCESS in 8m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/49daaee5aa9e430598e5396ae78820d5 : TIMED_OUT in 41m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186deb74302c4b9ba6c65a3fdc5f905c : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/840a31ac96434cf789f17a347b7ab062 : SUCCESS in 1h 12m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e17f24b96a604e69a4bfc32539a42d94 : SUCCESS in 1h 12m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/03f6ab887c6f47ac8682082988639bf8 : SUCCESS in 1h 29m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ee37627fb5e47a8a7e3743bbbd90e31 : SUCCESS in 1h 07m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b35a21abf1874fafbf3e9e5f62f5019c : SUCCESS in 9m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8a2419b5fe224cf9abfacbcfbfe00514 : SUCCESS in 22m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8b4e02de47204aa88222d61650fe4121 : SUCCESS in 4m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83f967a4609148cb919e83ac9ccc5084 : SUCCESS in 6m 40s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/adf54e4761ab4dd8aa8bb58a563638f5 : FAILURE in 22m 58s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/756884a5f1a84012be1fc4b739efc5d3 : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/aa0fc9b1b1cc40aaa039ec89ee2f9780 : SUCCESS in 7m 57s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c5a20b773dbd4de9a8c4e483169defdf : SUCCESS in 8m 41s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0bb5bfa04d1b4638827b30aca8d0ab41 : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bdd7f8f2aa2341c3ad95b3e95647425f : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9685be93304246f7a9030ed6951f7327 : SUCCESS in 1h 02m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c20cb14c9bf74fbc9c66b0819e16a1ae : TIMED_OUT Host unreachable in 3h 39m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0fb408384590429d8fb67489b0e72a7d : SUCCESS in 2h 09m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/96bc8bc5c50341e6ba36f22e8b1964e8 : FAILURE in 2h 20m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/24deb415b9f14aaa964014c5d79126a1 : SUCCESS in 1h 49m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/561f23c5912e4ed49ad3713a02fecb39 : SUCCESS in 38m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/660247cf6c71406daa853ed098bf0751 : SUCCESS in 1h 07m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/314bcf2703164442ab20cc70af873984 : SUCCESS in 1h 29m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a82bc13c14bc478d98417e3804c24082 : SUCCESS in 44m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/117e95a09e62462e9d9ce08a61a7f8aa : SUCCESS in 1h 56m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c6930a76e474726a649247c0209ecf8 : SUCCESS in 34m 20s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"77413e71283548d82a5a05d69be2ca4164e31ac1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-30 11:40:54.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_06_2026_11_32_59\nLeave a comment with","accounts_in_message":[],"_revision_number":14},{"id":"958b29badd91d2d911b431832ba0be0be7791469","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-06-30 12:26:14.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0206228d3314a7ee24da8f6ada403da6790d13bb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-30 13:27:55.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7dee3752f4483dd5a952ef2b944ab640ae810fa3 .","accounts_in_message":[],"_revision_number":14},{"id":"a264d82b6fab4cee700eedb72d2affe168a899a1","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 15:45:30.000000000","message":"Patch Set 14:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/390 : SUCCESS in 0h 41m 42s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/953966","accounts_in_message":[],"_revision_number":14},{"id":"04e7165b6d39dcae8d00ca78393624d689938147","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-30 15:52:06.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/fcffaa87595c409a8efa49921c491116 : SUCCESS in 51m 08s\n- cinder-storpool-tempest-iscsi-multipath build/5dde72032838404097f038985de79ee9 : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":14},{"id":"45cf0355e0c4152edb16001caaf638c210e9672c","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 16:00:17.000000000","message":"Patch Set 14:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/296 : FAILURE in 0h 30m 16s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/953966","accounts_in_message":[],"_revision_number":14},{"id":"00f26b5d943ac34ff8739f90b13f675bb8ba5094","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 16:48:37.000000000","message":"Patch Set 14:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/662 : SUCCESS in 0h 40m 55s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/953966","accounts_in_message":[],"_revision_number":14},{"id":"791b5338880dd55aa0f8c0934c1667a788df90a2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-30 19:19:14.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4585/.","accounts_in_message":[],"_revision_number":14},{"id":"2873bca9adbfa5734a29937f791efb1ec80fe9b7","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-30 22:30:18.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-66-953966-14-fc : SUCCESS in 1h 12m 59s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-66-953966-14-iscsi : SUCCESS in 49m 35s","accounts_in_message":[],"_revision_number":14},{"id":"2a8c2f0024ddb1930b8a4c45eeb07ec1f819b35f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-07-06 20:22:04.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/66/953966/14/20260706232357/NEC-ISCSI : SUCCESS in 5h 29m 48s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/66/953966/14/20260706232357/NEC-FC : SUCCESS in 5h 58m 04s","accounts_in_message":[],"_revision_number":14},{"id":"f3afa4932550729b466ec531b56539f6b0c5956f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-08 23:24:06.000000000","message":"Patch Set 13:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4219/.","accounts_in_message":[],"_revision_number":13},{"id":"1116d0cff7f032a1d0a680990372588fbd3b932b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 12:06:45.000000000","message":"Patch Set 14:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4409/.","accounts_in_message":[],"_revision_number":14},{"id":"48ed5a8ff4ffa1177c76b95c4a060b5bbabcb5a7","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 15:31:33.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0a065f8b55ad9945535e50c3c1acbb5e5383e421","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-07-29 13:57:14.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"898df2c04cd0d465c7b455ec4a26503853361d29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 13:58:26.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"44c9300fec7e64be3f179f3882b6f2f14d86e876","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 15:33:35.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78252998562a4efc945019ba4fcbf1ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fb09285e23448578f31fa8d77f662cc : SUCCESS in 9m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd0c1d1865f64525b942489e8e188f8a : SUCCESS in 12m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1cc606c751344321a01b3b6058527c4d : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/089c0fc4469240449273666964a4d421 : SUCCESS in 15m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/62fb49765c9a43a28c4aca903a457c88 : SUCCESS in 1h 11m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/64624d55412842899b6eb1bb43601338 : SUCCESS in 1h 09m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/124a189c6dcc4396af7937c19d28b0f1 : SUCCESS in 1h 29m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5373afadd69547f5b1be2519e21413eb : SUCCESS in 29m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc774d9cb5fe44918f8093fb31bb6ede : SUCCESS in 9m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c30e3e6e126c4fadadf79dc0fdf99679 : SUCCESS in 1h 20m 57s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6bd76dea4940473cbabd0d0d57d576d5 : SUCCESS in 1h 28m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f60bc26bc7b47c2b9c282f9f7450b1a : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":14},{"id":"f594ef63551f15891d45fa425ce8e03d2de333ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 15:33:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"1ccdfaf50dd4f2a73eb5526058ade54ad70da0c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 15:35:38.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af73faa6d25a43d08c41b50ccba23d51\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70089bcb4f95439cb90f99d8726db10e : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/754b5526f3fd4b49ab9e7d90d80ac246 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"57dc9ece9652af87535c533360acb2f07a8d5e0d","revisions":{"04ee51c2d1a8b24ed22340c8cc53feefd95f9875":{"kind":"REWORK","_number":1,"created":"2025-07-02 15:47:06.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/1"}}},"commit":{"parents":[{"commit":"53245bce31549a29ce74dd49de6c9f7417868126","subject":"NFS driver: Fix fail to resize NFS volume with snapshots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53245bce31549a29ce74dd49de6c9f7417868126"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:46:26.000000000","tz":120},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04ee51c2d1a8b24ed22340c8cc53feefd95f9875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04ee51c2d1a8b24ed22340c8cc53feefd95f9875"}]},"branch":"refs/heads/master"},"819cb4c0a540afb6a722024473703e5ea4d302f1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-04 14:22:37.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/2"}}},"commit":{"parents":[{"commit":"27373d61fe54e55afa91f1e93cc65d3dd0582f9f","subject":"Update PowerMax driver doc support matrix for Caracal Dalmatian Epoxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27373d61fe54e55afa91f1e93cc65d3dd0582f9f"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-04 14:22:30.000000000","tz":120},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/819cb4c0a540afb6a722024473703e5ea4d302f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/819cb4c0a540afb6a722024473703e5ea4d302f1"}]},"branch":"refs/heads/master"},"8b71cbf025e12e3871c0317aed6b75c6c8939556":{"kind":"REWORK","_number":3,"created":"2025-07-04 14:39:10.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/3"}}},"commit":{"parents":[{"commit":"3948daf0078e5b36a9afcea02dbc576ad0ab2183","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3948daf0078e5b36a9afcea02dbc576ad0ab2183"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-04 14:39:02.000000000","tz":120},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b71cbf025e12e3871c0317aed6b75c6c8939556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b71cbf025e12e3871c0317aed6b75c6c8939556"}]},"branch":"refs/heads/master"},"13532f8970316afddb1a12db73d9030bd6ebbf46":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-09-16 07:35:52.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/4"}}},"commit":{"parents":[{"commit":"f5036b85d11b2fa7d1aa87f3df3631acccacd4b4","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5036b85d11b2fa7d1aa87f3df3631acccacd4b4"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-09-16 07:35:26.000000000","tz":120},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13532f8970316afddb1a12db73d9030bd6ebbf46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13532f8970316afddb1a12db73d9030bd6ebbf46"}]},"branch":"refs/heads/master"},"7f4f4d68c1105bada3a5b0c48e0663ebaeb088bf":{"kind":"REWORK","_number":5,"created":"2026-01-13 13:43:42.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/5"}}},"commit":{"parents":[{"commit":"fc59321a9c0188fcb9fa389a9b67884a074e87de","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc59321a9c0188fcb9fa389a9b67884a074e87de"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-01-13 13:42:49.000000000","tz":60},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f4f4d68c1105bada3a5b0c48e0663ebaeb088bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f4f4d68c1105bada3a5b0c48e0663ebaeb088bf"}]},"branch":"refs/heads/master"},"80269225c386db62e91e011d4ee996165013f1d1":{"kind":"REWORK","_number":6,"created":"2026-01-14 13:24:13.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/6"}}},"commit":{"parents":[{"commit":"89612e1946f36218637b45587b2a8460ca7262e6","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89612e1946f36218637b45587b2a8460ca7262e6"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-01-14 10:53:50.000000000","tz":60},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80269225c386db62e91e011d4ee996165013f1d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80269225c386db62e91e011d4ee996165013f1d1"}]},"branch":"refs/heads/master"},"976f163b26194c488ce7f789d5aa4a8e29bdebc1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-01-26 10:11:19.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/7"}}},"commit":{"parents":[{"commit":"2d2124d972710f9b2ae615aed838ae823defb60f","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2d2124d972710f9b2ae615aed838ae823defb60f"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-01-26 10:09:58.000000000","tz":60},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/976f163b26194c488ce7f789d5aa4a8e29bdebc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/976f163b26194c488ce7f789d5aa4a8e29bdebc1"}]},"branch":"refs/heads/master"},"788b030c5e3ca75a7b501990fff4359b607f35a6":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-03-10 14:27:43.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/8"}}},"commit":{"parents":[{"commit":"cc6c6ed232f53e5c008bfbd9a8422c7132204bb6","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc6c6ed232f53e5c008bfbd9a8422c7132204bb6"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-03-10 14:27:08.000000000","tz":60},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/788b030c5e3ca75a7b501990fff4359b607f35a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/788b030c5e3ca75a7b501990fff4359b607f35a6"}]},"branch":"refs/heads/master"},"69b8f07504b431e6b052be16341c5e400d7fa0f9":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-03-11 16:22:30.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/9"}}},"commit":{"parents":[{"commit":"23f32d1c6f5e09059b7cb7c2d1137b9f3a03049e","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23f32d1c6f5e09059b7cb7c2d1137b9f3a03049e"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-03-11 16:19:37.000000000","tz":60},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/69b8f07504b431e6b052be16341c5e400d7fa0f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/69b8f07504b431e6b052be16341c5e400d7fa0f9"}]},"branch":"refs/heads/master"},"f2ef0ce341dae4fd03b6a1cc4159caa4f4370214":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-03-18 09:56:12.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/10"}}},"commit":{"parents":[{"commit":"6bed6db6c4bd8d65225edff575b6a5190bf5af3c","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6bed6db6c4bd8d65225edff575b6a5190bf5af3c"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-03-18 09:54:09.000000000","tz":60},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2ef0ce341dae4fd03b6a1cc4159caa4f4370214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2ef0ce341dae4fd03b6a1cc4159caa4f4370214"}]},"branch":"refs/heads/master"},"cc9cdfacb963df5a15b9d58e06157fd5206d7609":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-03-30 15:53:33.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/11"}}},"commit":{"parents":[{"commit":"955abe16f629b3c01f1383cb57c3735c8f2e0695","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/955abe16f629b3c01f1383cb57c3735c8f2e0695"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-03-30 15:52:38.000000000","tz":120},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc9cdfacb963df5a15b9d58e06157fd5206d7609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc9cdfacb963df5a15b9d58e06157fd5206d7609"}]},"branch":"refs/heads/master"},"b6186c981f2ed0ad3e934ae750adf2307079b4a0":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-03-31 14:24:05.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/12"}}},"commit":{"parents":[{"commit":"057452e5006191384694f4115403fe70091e7131","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/057452e5006191384694f4115403fe70091e7131"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-03-31 14:23:21.000000000","tz":120},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6186c981f2ed0ad3e934ae750adf2307079b4a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6186c981f2ed0ad3e934ae750adf2307079b4a0"}]},"branch":"refs/heads/master"},"590b7cc14d704f105de2a7d2b3bed35f866ba4e1":{"kind":"REWORK","_number":13,"created":"2026-06-08 13:27:15.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/13"}}},"commit":{"parents":[{"commit":"8669c50cbf45566a99f19f89451f3378247aba58","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8669c50cbf45566a99f19f89451f3378247aba58"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-06-08 13:25:21.000000000","tz":120},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/590b7cc14d704f105de2a7d2b3bed35f866ba4e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/590b7cc14d704f105de2a7d2b3bed35f866ba4e1"}]},"branch":"refs/heads/master"},"57dc9ece9652af87535c533360acb2f07a8d5e0d":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-06-30 07:43:14.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/66/953966/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/953966/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/953966/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/953966/14"}}},"commit":{"parents":[{"commit":"7895fffba0ae204d147b8827c93ef76b6cbfc536","subject":"Report provisioning in driver stats for Quobyte driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7895fffba0ae204d147b8827c93ef76b6cbfc536"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-07-02 15:35:20.000000000","tz":120},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-06-30 07:42:57.000000000","tz":120},"subject":"Quobyte: adapt to image_utils fixes","message":"Quobyte: adapt to image_utils fixes\n\nThis adapts the Quobyte volume driver to the image_utils fixes\nimplemented in change I49404e87eb0c77b4ed92918404f86c073fbfd713 .\n\nThe driver now passes image info data into the convert call to prevent\nfailing on fetching info from the backing file of an image.\n\nRelated-bug: #2074377\nChange-Id: I7dee3752f4483dd5a952ef2b944ab640ae810fa3\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57dc9ece9652af87535c533360acb2f07a8d5e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57dc9ece9652af87535c533360acb2f07a8d5e0d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
