)]}'
{"id":"openstack%2Fcinder~954223","triplet_id":"openstack%2Fcinder~master~If5f74fb6bebb6762d1db471e814b94d2374ddfb6","project":"openstack/cinder","branch":"master","topic":"bp/dell-powerflex-token-refresh","attention_set":{},"removed_from_attention_set":{"37792":{"account":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"last_update":"2026-01-28 09:07:05.000000000","reason":"removed on reply"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-02-17 14:54:29.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2026-02-17 14:54:29.000000000","reason":"Change was submitted"},"37786":{"account":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"last_update":"2026-02-17 14:54:29.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-02-17 14:54:29.000000000","reason":"Change was submitted"},"35759":{"account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"last_update":"2026-02-17 14:54:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If5f74fb6bebb6762d1db471e814b94d2374ddfb6","subject":"Dell PowerFlex: Update REST token refresh mechanism","status":"MERGED","created":"2025-07-07 10:59:42.000000000","updated":"2026-02-18 07:50:35.000000000","submitted":"2026-02-17 14:54:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":48,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954223-bp/dell-powerflex-token-refresh","meta_rev_id":"c232f0a257e3e06ab431f666365e4386244d6e70","_number":954223,"virtual_id_number":954223,"owner":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 14:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2026-01-28 23:33:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2025-10-28 15:56:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2026-02-17 12:55:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2026-02-17 12:55:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-07 11:00:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-07 11:02:12.000000000","updated_by":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"REVIEWER"},{"updated":"2025-07-07 11:47:09.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-07 12:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-07 12:37:14.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-07-07 13:22:10.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-07-07 13:30:27.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"state":"REVIEWER"},{"updated":"2025-07-07 13:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-07 15:07:11.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-07 16:53:59.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-13 10:26:17.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-08-26 19:56:46.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-09-15 09:04:43.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-09-15 10:18:56.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-10-14 06:22:42.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-10-26 15:06:33.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-11-16 10:54:23.000000000","updated_by":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"reviewer":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"state":"CC"},{"updated":"2026-01-28 05:07:20.000000000","updated_by":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"reviewer":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"state":"CC"},{"updated":"2026-01-28 23:33:03.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-02-11 06:13:36.000000000","updated_by":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-02-11 06:13:36.000000000","updated_by":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-02-17 12:55:50.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-02-18 07:50:31.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-02-18 07:50:35.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"9ff8ba9329450a8e94183361843081f7127b8daa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"date":"2025-07-07 10:59:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4dba53b24e1e220c584f700cc5aab19c4190440","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-07 11:00:36.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59053a02e51ec0cdd32c5d942d1ef845e86024a8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-07 11:47:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/954223 : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":1},{"id":"384304dc28e06d1f2da9ae55ab5c041db95a5bc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 12:24:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6fc0c8282ba41f49521551511a2d6c4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c20466684476421ca9934a4cf877210e : SUCCESS in 23m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e9d4c560db404aa88a5cccc8ce002d0b : SUCCESS in 33m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7093c4edfdc44b799b383141da6ad434 : SUCCESS in 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed6bc063e7287fe179dd493ecfbc1f4879d6a58d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-07 12:37:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1174/.","accounts_in_message":[],"_revision_number":1},{"id":"8f24d86a4c9df3355b571a67a870543b542b0e2c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-07 12:41:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1304/.","accounts_in_message":[],"_revision_number":1},{"id":"83b69ddabedd513a619b60da2eb0904d9c95b468","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-07 13:22:10.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4c4a3a74fea74807be32b3ec0b837f3b\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0c3697793b7f43eb9cda20c5616b0265 : NODE_FAILURE Node request 200-0000071375 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/93088c654d164dd695f51c2a7ba709c2 : NODE_FAILURE Node request 200-0000071376 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/835ac276eaf848969ea8b84a52f574e7 : RETRY_LIMIT in 25s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/074eb2b8ecba40efb82b2a418500a471 : NODE_FAILURE Node request 200-0000071396 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fec8a125bde7419889930edaae7a5f31 : RETRY_LIMIT Host unreachable in 50m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0538bdabccbb1d7f9f9511ab3f9fdc6b6900710f","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-07-07 13:30:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"aab24d1adad0ebef4a7efe5927ca90d60565c8bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 13:52:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0448420a900416290a6aa1fac0644fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d67889f5d0984428af1e4609c5e788e7 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9029cf67b3d3429880b142c13b80924e : SUCCESS in 8m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7fd6a5432a564b5c9cd04f9fd4c13c74 : SUCCESS in 11m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2fb463f34f846cf9e9c0b5563c36cb5 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d395ff7575e4590b83be02ab49722bd : SUCCESS in 13m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/f88358d392d3473789755b40dcb9134e : SUCCESS in 1h 00m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e640a940c9884067b20c54f24b6d847f : SUCCESS in 33m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e77c6b83c2184eccb53857834ec03480 : SUCCESS in 1h 17m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fff9fc9f012b4d48a378e7d4c51e5118 : SUCCESS in 32m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55e481592d264043adafbfb79b5c3134 : SUCCESS in 20m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/594b8cef16704052b65ee600e85ff658 : SUCCESS in 24m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/394895d086a44c9b866151abec9a9c6c : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/354a8b5d654744c085b718e1e2385426 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b0f5f354f1664f489283f1e02e567b50 : SUCCESS in 6m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/be15612da4ef4b548b76e9ec5d013e0e : SUCCESS in 9m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8a2aaf96dfa14323b31a96ccd4080111 : FAILURE in 34m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b9e1bfdad444d1d9c6f28c4346f26c7 : SUCCESS in 3m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/880cca3480ed412ea69490c6003bded0 : SUCCESS in 1h 16m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3ae1ddb878d641d89d43dc83e80bbd33 : FAILURE in 2h 49m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5dc7b47a1e0143e28db23ae7a189da78 : SUCCESS in 1h 47m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f1dadf95f9334ca98672dfbb9b47646e : FAILURE in 17m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9a37d1b5c18d4b4d9119da663bf16f13 : SUCCESS in 27m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ba5867c51d44497cab9c4b000a408460 : SUCCESS in 1h 16m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/03fd032b32c74c8da4da05f6b516f741 : SUCCESS in 1h 14m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/afe557ded38144a09a5f7b47abbcdfb9 : SUCCESS in 42m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9021d30f7232475caf0c3a7f3cc539c9 : SUCCESS in 1h 32m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0f3cffbb50104004b5c2c2eb16cd6bc3 : FAILURE in 17m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1646ca3d963d44caa82aed7b79cdde24 : SUCCESS in 1h 16m 40s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/452bba0a20a34e4babfde196942cb09b : SUCCESS in 1h 29m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce53985406e34a9db16c082128f5b383 : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3c5c018748c947d51acf9eeb41e29deea9f3b773","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-07-07 14:11:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a92b30a99eb079a761142eafb27820bdc504be0c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-07 15:07:11.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"11635bfd3bdd82d71ed094858c79840bd06649bd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-07 16:53:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"55021015e7db507a3e76f7dc8b61c76688f5ceb0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-13 10:26:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/23/954223/1/20250713133915/NEC-V-ISCSI : SUCCESS in 5h 15m 54s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/23/954223/1/20250713133915/NEC-V-FC : SUCCESS in 5h 47m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2a2d1de6e9acfc1135d0c64020702db2e551a634","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-08-21 07:52:34.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cedec28a8dbd626da23a2fa94c61d05ffc845b67","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-26 19:56:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d60faaa05aee6fef0ca3c51f70b1606f150f88cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-15 09:04:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f9d960488de97b76b14e55d68947b9eb3ebdabed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-15 09:04:43.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/583/","accounts_in_message":[],"_revision_number":2},{"id":"14756d806de294244defeb39bd927a870a64393c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-15 09:05:56.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"a561a57c3e53090219cb7f311b7911c43bb04b85","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-15 09:11:51.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bd3597d6acb161e4c6702d3118977b9d12d4acfb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-15 09:23:10.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/954223/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/954223/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"91b31c22233e6a4505a84fab22c8506dfaf52ec3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-15 10:02:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/954223 : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5aa7722b949b7286288fac370778eaf076add4eb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-15 10:08:06.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1794/.","accounts_in_message":[],"_revision_number":2},{"id":"a58f9911a3c8d61d55e1ac7a8dea81e9333d4363","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-15 10:18:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/954223 : FAILURE in 1h 12m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f8aa7de8fb6be7d629d882e94b9c63dacddbdf78","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-15 10:20:23.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1928/.","accounts_in_message":[],"_revision_number":2},{"id":"8d61ffb7798bc60e97ca35549d39f99942989b6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 11:27:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66f1b27442a6438e86a36d95bb41ac2a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38b7a47a355445cd9e564be8fcbc2882 : FAILURE in 26m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f441149aa8c4d889beabdd6093a3710 : FAILURE in 23m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6d1e0870047d4d08b08c330cc28af134 : FAILURE in 31m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f639afac7993c0af2c4749c341f91097d6e2a719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 11:44:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3095117e8b6a4ec2b39b0285b5f098fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962a4f5415c64b168c3a6b18af350755 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a67b72376af4d96b09870c238acfe46 : SUCCESS in 14m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c9864e2c0b54eb7b7fa064b509dfb00 : FAILURE in 13m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b7106a8a3fa493b8f8313067c9f67f3 : FAILURE in 15m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e92448dbf9d4439b7fa47b226439e78 : SUCCESS in 17m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/08ca535a98e9420188ca27ed6f5ec3f8 : SUCCESS in 51m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/afcbac3730b64b9d921c611e223d544d : SUCCESS in 1h 00m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45439b34debc4058a00e5dde9acf425f : SUCCESS in 1h 47m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc6da45148ea4136829d6a84efbbdc26 : SUCCESS in 53m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e33f0e12d804ecebe509a8d8ebea3ce : SUCCESS in 12m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a497bb7b180247b9a77ab9aa0e5d2372 : FAILURE in 17m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a8131691932a4d079aaa6a59c95c49e0 : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5e8f5ce7c56e4d50a6e70681aef3aa01 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/96e846ebe4514c618eeb84a76df7b113 : SUCCESS in 5m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1b541802e71841d9bed1f1c01ccd44ff : SUCCESS in 8m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3a80a3a4c2724dbbad688923b73db679 : FAILURE in 35m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9b7a951083344581aac2912d4bc700f5 : SUCCESS in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e1e83f4e382446e2a42014ac893d3b74 : SUCCESS in 1h 49m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/198cc4fc70fe40a3ae275fd98e0e5b91 : FAILURE in 2h 24m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4d96ed5c68e14521b5fa1aa5ca2e72fb : SUCCESS in 1h 53m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3f5d3abe75714dedbb2bf5ce9a8ccff9 : FAILURE in 13m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/50d8a3033d714610aabd9ae27e7c465d : SUCCESS in 37m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d7e5f73b91004d939266d52b9219d4c9 : SUCCESS in 55m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1500f2f8b30e432091a23b16b66b559b : SUCCESS in 2h 08m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6c2a2e927686407a8f29cc48247c3c10 : SUCCESS in 46m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/41f5f3b499c34e55a4738c5048cb137a : SUCCESS in 1h 42m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/81dc2a70ef3e4c7bb2f15fde107b026b : FAILURE in 17m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c94dcadec42b493ea4450bf6632bac72 : SUCCESS in 2h 04m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fec6342847a47d4bb0eeca1c54a1c1d : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":2},{"id":"97098318414111588d4e712b558342a4a627b582","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-15 13:28:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"31bd9ca4ba3038b8c854903d39c98793566baa54","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-15 13:29:03.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/584/","accounts_in_message":[],"_revision_number":3},{"id":"e1fb11083ead17a90bb8edc9073f2884fddf7093","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-15 13:30:23.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"0d8aad0d1fcfd30a2f539058bf8c3837ca3a6c1c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-15 13:47:40.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/954223/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/954223/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"ec36a7158c34a87158c15c7218da53d043e24bee","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-15 14:25:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/954223 : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":3},{"id":"4e99405e88178e67489f75cd7b3da3338882b4d9","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-15 14:26:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"596fda136f4c77de08c6f4c5ed2e014e7ee2181c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-15 14:33:15.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1795/.","accounts_in_message":[],"_revision_number":3},{"id":"5975fedb67770e25005fdc041be61f8f11cc1099","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-15 14:39:04.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1929/.","accounts_in_message":[],"_revision_number":3},{"id":"af6b1948a27a17a24917c708d722c32199795f12","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-15 14:43:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/954223 : FAILURE in 1h 12m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ef50746d4fa9f4690df518187450aa1bb96bcbfb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-15 14:43:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ca946ea125b64bfea94d96316970fe19 : SUCCESS in 48m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fdac9588816844d68ea312d03915abc9 : FAILURE in 1h 01m 17s","accounts_in_message":[],"_revision_number":3},{"id":"f86e693ca36740085ab0c7a42ca2e08ae302fde3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 15:45:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/940af570bbea49a98f3810b9174952e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ae347657fd44109a07e115563b0c9f : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddfcaa835f1a435a9272ad7e0ab425cb : SUCCESS in 11m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7902e34f5a5b432ba421ce44c2b1654f : FAILURE in 10m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8b0763c80d749b086e4d3e4b975f812 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde1d8f338124e8f85efe29a167c2a15 : SUCCESS in 11m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/85c5fd65c59b47a0897609cefc620a1e : SUCCESS in 46m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c1f5ce653c0247a3877b350dd77b79f2 : SUCCESS in 47m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6b7c9580fac24f769ef14c6a0a6c8432 : SUCCESS in 1h 40m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f6492429fb84b1080fcd1135e7e527e : SUCCESS in 1h 03m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f9552aa791c4f6293286264e75f9bd2 : SUCCESS in 14m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fa0d9dd9bd3444f487f6b760c462d98a : FAILURE in 21m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4ede6aa16b9c40a58a795826a639aee8 : SUCCESS in 6m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/214443119f7346969dae9fc5fcd377b0 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cfe25f71136b48d1bbba76125b2de1e6 : SUCCESS in 7m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c90330650b684f03b720d365fc460089 : SUCCESS in 11m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7a073df1af2e4eecacf8507519068bba : FAILURE in 37m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f09e07b0da4c4ae1987e249b84d60788 : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e33056690564f128a74d48c7977057c : SUCCESS in 1h 50m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e56d6d7926d749ba820d1c69558ac76d : FAILURE in 2h 13m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a4ad563ad8634d36bf0675b7e44aad69 : SUCCESS in 1h 41m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8c13be75678842969d8000d179546ed4 : FAILURE in 12m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3bb0671c0d2b4fc1be24a8401ef63352 : SUCCESS in 34m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0fc8d50dc7ee4ef489eca1042662b93e : SUCCESS in 1h 06m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5008a3c5e3934229a4e6948871ea541d : SUCCESS in 1h 27m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0c99600e71f94bd9a0ef4e96631e1e71 : SUCCESS in 40m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/729213a8d4214c3aba7704a0483625c1 : SUCCESS in 1h 24m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/51e14d2d44964e24a894fc0d5a9c416d : FAILURE in 19m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d944eb40e3a74d3086f754f5892b7205 : SUCCESS in 1h 44m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1458a3967214b79a414a566825482fd : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":3},{"id":"d5e38fd27e8b74581b46edd97941ca061f070b02","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-15 15:49:15.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,nlogs at https://oscilogs.quobyte.com/logs/If5f74fb6bebb6762d1db471e814b94d2374ddfb6 .","accounts_in_message":[],"_revision_number":3},{"id":"459a287d5f38bb6be9cbe14f16fd7e60e1990094","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-09-15 15:55:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"91a61a8abc516b5938f3b92ed372b68bacbc6d2e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-15 17:06:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1af6e897465f43bea2404fbf23eac385 : SUCCESS in 45m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/01efa79a816e472d81a62813af7af874 : FAILURE in 58m 07s","accounts_in_message":[],"_revision_number":3},{"id":"95c1f4b2278a7c95eb4eb8c92e32d7149b768d35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 17:38:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a21b557b27ef43cc9f8e8d4e0a519aac\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/782b1ad442eb4136a866166f1e55f5b6 : SUCCESS in 27m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a074bcb8a8fe413a904908c8bf9d3bb7 : FAILURE in 28m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da62e2d760174b95866a2ddfb43a5275 : FAILURE in 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c653ede641ec766a76aed32d2fa469dd88cd429e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 18:40:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5208d7f14734ca7a3c0886508646f00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffe30098fc1f4806b15af808c64de4ad : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/983abe9f3c104fed86ecf15f75bcdca6 : SUCCESS in 11m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e69dc37d99284f83bb95a34c664a1da4 : SUCCESS in 9m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cae8879669104075aaa2cd266e6216ab : SUCCESS in 12m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409c12ab8f034df2b38c943df42dad9c : SUCCESS in 14m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/28fbaab133f4468bacbe6b723dc0363e : SUCCESS in 1h 01m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ab85dbc10994669ad796a8c08b568d3 : SUCCESS in 1h 00m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c8a5be67722402cb97d7db8fdf1964e : SUCCESS in 1h 20m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f6ed394d3dd4519925565c7cb1ef65e : SUCCESS in 1h 02m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a19aa6381fdb40caa143c8a998138523 : SUCCESS in 18m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/65330145043b48c99b3b2ee4330b6f41 : SUCCESS in 19m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/65a819b6dc3843bca7b96c3f43166891 : SUCCESS in 6m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/412954ce100345f2b244610c02f54051 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0c21691d3c594aa2b02e2b06276371b2 : SUCCESS in 7m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a966c0b21b714b6abb00b82da2f6ef3d : SUCCESS in 8m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/653bb9d1969149ffb302f582bf7d9d3e : FAILURE in 39m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8be2d567aba24829942cb95a2056d63c : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9bc3832dcb7e42c984277ad507ebd101 : SUCCESS in 1h 29m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f51a274826974365a1cea5c768188981 : FAILURE in 2h 33m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8b513674f5e44c62afaed8204aecb0b7 : SUCCESS in 1h 17m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/312ad2cc7cef4c3f836caa882aeae3f1 : FAILURE in 9m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/207eebf00915438095952030653ce080 : SUCCESS in 25m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/65551d1639e1421da1c1fadce54ef0ec : SUCCESS in 54m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/46e58b5eaaf146c893cae71db542f271 : SUCCESS in 1h 45m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dff2b3ca65fd43c9a07c755899f8c964 : SUCCESS in 29m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/33edfbb782c34bd8a752872874c4e8a2 : SUCCESS in 1h 03m 25s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fb9336fa069a44a4b1d2a4e7dd27a400 : FAILURE in 17m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a3c86d38df1436e9535fb3c934fdaf0 : SUCCESS in 1h 58m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/648f2fa16dee4ca0b52e6c49a2352206 : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":3},{"id":"e454ff385ba6efeff413f941126d79005f8cc46f","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-09-16 16:44:06.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9d8ed2a9664cd6392db05e9843943bcef1c3f9e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-14 04:06:39.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2901793a3da177dd959517ddb74ea8dd4e8fbb7f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-14 04:07:32.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"878e3219a2810ac10cb7395ce14b4b374dc73a7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 04:35:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d843fbd97e8e4fbab140695b22e43d4c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/43b3cda3ea074fde926c16d0df7051ba : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ed89add4986486a864932aaba8e9792 : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8ccd55f90ca040dd84a4c4e87877b289 : SUCCESS in 26m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6e86ee0b48fb2da9085c12d6fdbc614dfa27fe49","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-14 04:46:00.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6ca99ee84d6a5cf03d19c5bde8223db240f7e123","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-14 04:59:17.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/If5f74fb6bebb6762d1db471e814b94d2374ddfb6 .","accounts_in_message":[],"_revision_number":4},{"id":"de9441dfb4389f4154efe2c6891d776c6986f47c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-14 05:00:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2136/.","accounts_in_message":[],"_revision_number":4},{"id":"823667c49fbf712daa891520cad63578c6578d53","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-14 05:08:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2022/.","accounts_in_message":[],"_revision_number":4},{"id":"bd28c6a42174fc57b61fccf470be8f3e615198ef","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-14 05:17:57.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e456fb7743a6430badde7afb5aadce83 : SUCCESS in 46m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/697b1511c8e740c486a4ff916a04a2ac : FAILURE in 1h 08m 04s","accounts_in_message":[],"_revision_number":4},{"id":"625dfacd7710c4701ea8e3fcc0f394e5eaff8fe9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-14 05:20:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/954223 : FAILURE in 1h 12m 05s","accounts_in_message":[],"_revision_number":4},{"id":"fbf4e6199cd43cfbf10aebac684b3b6e6cb7849e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 06:16:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4487cc797114865abb9291acc9482b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaba6a7cf34c416086193cffdd7ab145 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09e6c54ffaba472092eb2a9f32c448e0 : SUCCESS in 12m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35ecd234ceb84e2c9084c9bd2779d145 : SUCCESS in 11m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b9c6ae503674c4994e30ec8225d9b79 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6498bddf9b44cb786b8577e211e86e4 : SUCCESS in 14m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/107e89ae85264ac5b2013165ae21fa50 : SUCCESS in 29m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e63ea0a71c0248b88f6d028729af2f01 : SUCCESS in 1h 03m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7505b80c992344c99631a17d281bb35b : SUCCESS in 1h 28m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8001451cf7f4580945cac356d7bb9ee : SUCCESS in 1h 01m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef86580c36f442008503be5d14669759 : SUCCESS in 15m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f4cc0f45b78e4fa9a6e3adeba4c9fea0 : SUCCESS in 23m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1a70aa4c251e4b1fb05a84475a6f35f1 : SUCCESS in 6m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/006b2340730e4c10a8c2e62b5af03ace : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dabbd35ded874f9cb663d4f99d73584f : SUCCESS in 3m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ef7e3a6a6e3349a4ab322cf2b653739b : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/023a35c0910d42a593914aa0c120a782 : FAILURE in 35m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3169c8a4fb054abca539c0e00d4fb272 : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6e65ef9d696749bc87954a06912b17cc : SUCCESS in 1h 18m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/640f69d5f33d4e1fa05ee7eff874f001 : FAILURE in 1h 35m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c9b926a6ddbd425aa1264fd630a814db : SUCCESS in 1h 54m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/59b97972c1aa463098aa5544fdef3dd5 : SUCCESS in 1h 36m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d1b076fb0edd45cf8d22612096a36604 : SUCCESS in 29m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bc48b0faebe64d829b8f3ef03caf3880 : SUCCESS in 1h 02m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/98cd3e8b2b1843aeb47ae3ec2309e87b : SUCCESS in 1h 30m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7015f7e7af6a4bd588ccb2a386755d8f : SUCCESS in 39m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/369920e0bfe94fd7867d0872031c176c : SUCCESS in 1h 11m 40s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/559d59a8bd784357945f663be70b7d3f : POST_FAILURE in 2h 08m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a0fc3f2b0cc4a00a086e1d6512b9729 : SUCCESS in 1h 54m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7049a4a089d7483dafc477cbd143cb0c : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":4},{"id":"60cfce3e7cf8843a6fdb999cd574a6aeeac54acf","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-14 06:20:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd37d35f269a87cdfc9a16ff3b52bfbe26550105","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-14 06:22:42.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_10_2025_06_16_13\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"82fd5ae8de615858e61f9a17886d09e3a3296d4b","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-10-22 19:23:27.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"41842c948292fcce1182843182fa415ec661a16f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-26 09:51:33.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7e27f6a4b15a33ad6ae454d6d728bc1e19418d92","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-26 09:52:22.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"eb46bc882b2661728658e0d4d7dd7cc3dbea7553","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-26 10:11:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97c115cfd9c34b22b0a1e75910f55854\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/529981dec890452d9ff5a80cca5ecadc : SUCCESS in 16m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fbcdad6133174f5190fdeee23b3e0895 : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/89173cd3de5e42cb98d43b1037989377 : SUCCESS in 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3f029062963b192f853c5ca8dff766f83cf7ce15","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-26 10:50:30.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/If5f74fb6bebb6762d1db471e814b94d2374ddfb6 .","accounts_in_message":[],"_revision_number":5},{"id":"0e486aa198bd7848010d4ed689412af471197798","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-26 11:17:01.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8c27cc1b5d8e4ff39cf984268c172794 : SUCCESS in 46m 42s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ea81714ffc0b4ce19c94e4f109e5c02c : FAILURE in 57m 38s","accounts_in_message":[],"_revision_number":5},{"id":"22dcb0101018c7d0bd7f07b0f417e07bade4940f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-26 11:22:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2179/.","accounts_in_message":[],"_revision_number":5},{"id":"36b130f3bed65b12c7ec779cc771060c2dac85f6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-26 11:23:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2293/.","accounts_in_message":[],"_revision_number":5},{"id":"973ee05d8323a4daa814270c3e62fdea3f05290a","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-26 11:25:12.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"71a90ec5b7514a6c83ae7d658040ede22152cea7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-26 11:45:38.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/954223 : FAILURE in 1h 11m 51s","accounts_in_message":[],"_revision_number":5},{"id":"406f332c7ec4ac2e2a090bdbfff1391a7f744179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-26 12:20:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2756d498f81a44ab9b53d04cdc90f7cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dea61a25ccb147f48bef82d5c684ee79 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c45c48812ef477b8aca2668df119f88 : SUCCESS in 7m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7136f8119ba041e9b37c476c5abfebe3 : SUCCESS in 14m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b404d84bd74416da6546668bf609e0c : SUCCESS in 14m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/e5be5a743cfc42e5a9eb7d2bb7c092d9 : SUCCESS in 58m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d88c87631e940aeab9aaa20f77eaca4 : SUCCESS in 29m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8579da599524283b86689ffe35b5b51 : SUCCESS in 1h 27m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a817225d7e864f9b9f6cbf582f1d7288 : SUCCESS in 31m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbbb1be658a6451bb52b709b1a52434d : SUCCESS in 10m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1b8b3c4398d84923a56ccd553a09d540 : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b94763fcc4564f8a869a06956e1f5c75 : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d2b3295945e44234b9ac851e004e8f14 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5421830d61f54051b035ea0682f4678d : SUCCESS in 5m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7fba3e8ddd204cc4acf81a1f479e78eb : SUCCESS in 10m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/03b7883508f84d33b0a985a988ec24b3 : FAILURE in 36m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89f21b4549334d42a2bfd6bda3b26d74 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b82319d7e5c64df7b841c0b884fcad62 : SUCCESS in 1h 56m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3f832a7d437d40f5bad169d7ca6b8462 : FAILURE in 2h 24m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/88cd660be8ce4ac2a0caa0919d426ad3 : SUCCESS in 1h 48m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a4f29d3a675f4e57a33aa14ac55638ce : SUCCESS in 1h 23m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/964033bc33904108ad8b55c3bae2f749 : SUCCESS in 36m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ed91400428ae44cc954b9efe51c033ad : SUCCESS in 1h 10m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f87a1b2963a8441eb67dc4fb853b1c7a : SUCCESS in 1h 35m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/146761fbc6bf43ae8281dcec0e993c17 : SUCCESS in 45m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a52fcb0c5b9a44a8a2528c27da5a4d8e : SUCCESS in 1h 50m 08s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2357512366a54989905ef32671398a40 : SUCCESS in 2h 05m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5680e41fb1004f4381560352d937bfaf : SUCCESS in 1h 32m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bc075e576924fd2a48b1b432cf38df9 : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":5},{"id":"5b9c4c1980bc7c01fc8b6a15b0fa3f9a32c538ed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-26 12:27:34.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_10_2025_12_20_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"83c448df8195ce983650ab38bdfa32b764bd3f56","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-10-26 15:06:33.000000000","message":"Patch Set 5:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-23-954223-5-fc: FAILURE\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-23-954223-5-iscsi: SUCCESS\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":5},{"id":"b1a62c5522180dba313c7277b17b64c37107b19a","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-10-27 16:57:20.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d573f93254bc7bacafd4501c8d11905eb57c0b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-28 03:19:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"999ccd4da0176571a9ee5732d574d4199399ee95","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-28 03:20:28.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"8cf7be042b468e93e1302038b32a5a45249c7d6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 03:41:20.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78fb9082955541158f2bb532d202aa4e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1865d771eee4d6a8145f9421bc548b9 : SUCCESS in 19m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2584e69fc9bf4eda80670fbc6886d38b : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d9edde464cc24f4b9b1f911f8f11d19e : SUCCESS in 16m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9a6b69c4164bfec87e2db8346b7250529542259f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-28 04:18:28.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/If5f74fb6bebb6762d1db471e814b94d2374ddfb6 .","accounts_in_message":[],"_revision_number":6},{"id":"d9299ebb53c492fcc14d53eadeda0dc23eec908a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-28 04:20:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2190/.","accounts_in_message":[],"_revision_number":6},{"id":"43d3c0d606da75963292d91d0b23d8da27bf3fb8","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-28 04:22:10.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d5988a04219858f588d8c3470fa0147495a985c2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-28 04:26:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2304/.","accounts_in_message":[],"_revision_number":6},{"id":"418d74aae143eccdb80adbc7ded48644b09d0afb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-28 05:02:13.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7af8896a475542b59c7f26e0b353f2d6 : SUCCESS in 1h 37m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2eb6e8282a91434aa0d1057879b4ac4d : FAILURE in 1h 39m 25s","accounts_in_message":[],"_revision_number":6},{"id":"c766e7e6a29240c8aa10617dfc420fdcc86156f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 05:56:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bae3f228f3ab42ab9c01599d29ab635d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b507e68e78614c889c2af6aede1d46bd : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85141c77beb0455e93a36c7d1ecd0bc1 : SUCCESS in 11m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f8b04de23c04a608570106efac5e9f2 : SUCCESS in 13m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a175e3650d6470693d8d9293206343d : SUCCESS in 13m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/4d3b2cc09598472eb0e0fad02c69738f : SUCCESS in 1h 17m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a02329715e874b3cb8a5a4c8b2a52bc0 : SUCCESS in 46m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c8f2db1a5a845518c710205a0f89ea0 : TIMED_OUT in 2h 06m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec71b590b3584031905af16075acca10 : SUCCESS in 32m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79325a38e75e4178a4b5794cf9da2c5c : SUCCESS in 11m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5d61b8f050014edaabffa5063973e091 : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a1c79b1d1f584a1ab02f61fe06739eba : SUCCESS in 6m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e58c2a8fa0b147e59db483dcf9abd7d6 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/97c754d979104bcda64c2eb9c3dbfa1b : SUCCESS in 6m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9e9fd567aa694744bb59d6feb3fc8277 : SUCCESS in 9m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4af8a2e78b0c44b2863fd58cc34f88e8 : NODE_FAILURE Node(set) request a60bfa68be6d466da2086cb42ecaac75 failed in 0s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/92c23eaed24d4a73a53d9011bbd1143a : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6d25e43e1c154762b9242d456f9c997f : SUCCESS in 1h 54m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d96b08071b214d5cb994268d287da8ad : FAILURE in 2h 31m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c4a180a206834909b913b941e44a27a9 : SUCCESS in 2h 00m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bb1750669f6f4f54a921457e2ded82c0 : SUCCESS in 1h 57m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bab46279c78f4842ac518c6775459ba5 : NODE_FAILURE Node(set) request 819fcf2db7704682afb521f4a2613cf8 failed in 0s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/629c2a9bb6984507a67a19071404e2a1 : SUCCESS in 1h 07m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0e43df35588a4f1eb44ca965236c40d0 : NODE_FAILURE Node(set) request c27f26f7fceb40539e19bbb0c3bd7713 failed in 0s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b91fdd8d074a4889a6f43b618dea9ddb : NODE_FAILURE Node(set) request ff2ccd51517742b9b937527ea3fbccba failed in 0s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/232d476da4324dd28f006972c393b5cc : SUCCESS in 2h 12m 14s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/883b03bfff874846a6850079d919324f : POST_FAILURE in 2h 08m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2d89d893cc0498698bdec7935aac46a : SUCCESS in 47m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dba34c5701d4c118d3068185ccf09a3 : SUCCESS in 1h 17m 17s","accounts_in_message":[],"_revision_number":6},{"id":"82fdba2d6b228c702589404d1001d73142b9155b","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-28 06:02:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"73f5e757d2fc6b17d71d46be68baec93c0654f77","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-10-28 06:12:49.000000000","message":"Patch Set 6:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-23-954223-6-fc: SUCCESS\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-23-954223-6-iscsi: FAILURE\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":6},{"id":"7db1b9df13e81a13653d599580cd1b03f1317a15","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-28 06:21:35.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/954223 : TIMED_OUT in 3h 00m 36s","accounts_in_message":[],"_revision_number":6},{"id":"3761286a4a5fc9b9fd822532d5b441c0a04fdb64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 06:24:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dd7dc7c68d0418b8470ea2ff0ff579e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ecd5e5d279b148bd9d108fe003882536 : FAILURE in 19m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81dfc77da5b147648977e8de2ccdaf69 : SUCCESS in 14m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd85504573e140ecabf06ecbf95e2a7c : SUCCESS in 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e882be5cb5ba4d3eede5622e6fdab7fc32bbfafb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-28 08:15:48.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/89949f114c3145f69a620cc94f727ade : SUCCESS in 2h 08m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9ab653b8e18549b0a318c1a77de37f1a : FAILURE in 2h 10m 10s","accounts_in_message":[],"_revision_number":6},{"id":"00d735c75937f8809c80266dcc4d005ca84626c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 08:52:21.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac38d2273cbb4877a341f5605a834d05\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb512a4a3f3e4fc582fe0c65f16254f9 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/781605fb630e4ff698b518895bf6fe92 : SUCCESS in 10m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1caed77eeb7540b5b7b539d61e849b69 : SUCCESS in 14m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a3699c200547988324e071072af5d7 : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/f4f126e1f9cd446a9bdbe1e5d44b470a : SUCCESS in 59m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ca6097691d6f49a88ea850ec14ea7dc5 : SUCCESS in 59m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/71b2772a22c2494394a5711cd5194be3 : SUCCESS in 1h 39m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5bc5c4ceb69a4bb7a78012297f3fd5cb : SUCCESS in 1h 05m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10b64a2b5997492887b3629dd315db5d : SUCCESS in 16m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cb2315ec11f74742b6ca23326351b5b6 : SUCCESS in 20m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eb296170119846619d83b51e65093353 : SUCCESS in 3m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d5ae8b1b087472aa776458a7da20940 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/16fd03762db64a39a4441264ba12ebf7 : SUCCESS in 5m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5e6220177bd246b1acbeab04c71015f0 : SUCCESS in 8m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13e0fb363e604159b610c57175e0a65a : FAILURE in 34m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1f2971c4d6cf4109b4c89b39a243e306 : SUCCESS in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/32b5573d30324a5386e6753cfbdbb090 : SUCCESS in 2h 03m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3274a73206b645f6a4acd81e476a9266 : FAILURE in 2h 12m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c1535c0e48e743f5abace63baa6876dd : FAILURE in 1h 59m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bf1f7db4b20140ddb0a00d84fe0f6647 : SUCCESS in 1h 33m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/796109e3594d4f75bd96fc930902c04b : SUCCESS in 16m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d3875d282db143e19a61f8c66f34cbb7 : SUCCESS in 1h 07m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f0767370fc784c479dbd50382abe2c94 : SUCCESS in 1h 03m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0f948678518e4a528b15a65a1bdb6fe7 : SUCCESS in 42m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b44b287a55114967981edb1d9bdaa5ef : SUCCESS in 2h 17m 34s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e96358b3c1f44e4aad8e8eac68373e8c : SUCCESS in 1h 57m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/58f1d10df19647f0a663a033376d8091 : SUCCESS in 1h 27m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89b9dd040e8042c982534b756c2a4251 : SUCCESS in 57m 05s","accounts_in_message":[],"_revision_number":6},{"id":"48bb18396af20f7784c3184a2aae0a5cc39fc8a8","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2025-10-28 09:08:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b5ba2b61669e0f771a850c20f8709b56b738b905","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 09:43:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32a529ac677d4282b80212007eed03aa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a7cc812055a049a1826a483c5ae7a856 : SUCCESS in 20m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ec04770a0d545f3a8a09df0dec30382 : FAILURE in 32m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f612bbcfa97e4d209fd29c01a39fcd78 : SUCCESS in 26m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"156e154dc17505e3bb84d4aa759f7b1095ff2517","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-28 10:03:59.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9eab7585549c46b4bbe8c0ef8e3a0294 : FAILURE in 46m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/34856cceca994c40a9799b95125f6b29 : FAILURE in 51m 46s","accounts_in_message":[],"_revision_number":6},{"id":"ffac963365e3d995ed37139f2c43c3dfb5dddc2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 11:27:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a7e4f865cd2490a97357411453e6628\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ade624e6b8e04e649086c968922926bc : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af2931af1b4840149e44f3a5e977107e : SUCCESS in 12m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/175f534b6a5849ef9d208e7c7ef120c6 : SUCCESS in 14m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7784afd93c1f4b72a39c723948cb8408 : SUCCESS in 15m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/19da847aa03c4b89bf9cd2f8d39cf46f : SUCCESS in 30m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0e30d338aaa489883015c1a54bc8fff : SUCCESS in 54m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/065724a906d44d2283b7fa08ce82a853 : SUCCESS in 1h 45m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/359057c7f513491d92fcb7f75d5f6287 : SUCCESS in 32m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ebfaaa3456644659cd15bfdd6fa2143 : SUCCESS in 9m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/53216d410c3b4146a31d48d3ec0b8b5e : SUCCESS in 15m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c580c7056da0402d86f28a828b552997 : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/61dcbf75c8e3404e874b3d778a7aa6bd : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/32f899d731544f6fb42b895cd47ec451 : SUCCESS in 6m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4bb1837253b44667822df6bd4a8dc8fa : SUCCESS in 8m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eb0962bf5fae4b7aa0eb9e473aff4304 : FAILURE in 28m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b2534f393a2c48e78b73d18a9076ce05 : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9c654b600dc140aaa40faf73d628ffb0 : SUCCESS in 1h 55m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ffde9445249543a88a832b4b1cb22817 : FAILURE in 1h 55m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/53c6812d10d244f3841533628f57d2c4 : SUCCESS in 1h 55m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5f12709915c348899edab44a49133730 : SUCCESS in 2h 10m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3f417d3715104aa480992b3378dd3f6e : SUCCESS in 35m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/405e0979f85f4552b7cc106a91662fa7 : SUCCESS in 1h 01m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/096a66eb5630433891718bf861ff2035 : SUCCESS in 1h 32m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/867c8647262e41d5a990fc013afe7ca4 : SUCCESS in 40m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d8cec9b2e08041f99dc06eed2f797e05 : SUCCESS in 1h 05m 00s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/da972be4768a4c86873dbfde4f8fce18 : SUCCESS in 2h 02m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ff195ec104784d8ca89ffae86dfa2893 : SUCCESS in 52m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ef623efcbdf4227834912a2a2b343aa : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":6},{"id":"0339f0e37450a5c8402e18f55c7f164043ed96b4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-28 11:34:02.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_10_2025_11_27_11\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"3d99bc9d9f6e661376854e9db2b082baa7ec4db1","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-10-28 15:56:27.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"69f6ddc543d9d0991c4936559b4163d8fb99d0cd","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-01-28 09:04:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"98f9153ef9177ba35a413f0432a6b2e1e737d0f5","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-01-28 09:07:05.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/c35122d4cbbd47ba8d81b7e31e170810\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/33ac0f40542c47a99bf729ad4751cfbd : NODE_FAILURE Node(set) request 200-0000048236 failed in 0s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/b0b1f21a0eff4f4d9d03f06e051cbd28 : NODE_FAILURE Node(set) request 200-0000048237 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"8d67fc1be6f2b30897701b78e744b143241be8f9","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-01-28 23:33:03.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b7e82729d054602a4a042c95f508a94d3e8bd9ba","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-01-29 05:54:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"451bcbd2a32a86db948531b1f595fd1af91a3401","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-01-29 07:33:17.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/27d13d7495fa4c43a8e1b782efc66849\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/061455b863b24494b8f73246e80abeb6 : SUCCESS in 1h 38m 50s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/b78561de6bae4e2597f507f42a730e11 : FAILURE in 21m 02s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"72096828532b899effd2ed95e07a738c5307e835","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-02-06 05:05:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ff55b2534f9c1f5e0b09d8c882b555e190301d5d","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-02-06 06:49:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/61d8ad72d15040f0b66fd057e20bbe3a\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/3195e2138ec54964835349abfd0aa596 : SUCCESS in 1h 44m 07s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"b18f483fc1ff46197256d40a134adc4b128d0149","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-02-06 06:54:02.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"aed260843ec43241e29fec7c9865b76a52ead0ef","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-02-11 05:52:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"27b676ce72ee6ea19bd4a2dac7a8da9571887325","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-02-11 06:13:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4e94ae35d4d5f2d9eb9c240749f37296dd9dfdaa","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-17 12:55:50.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"051d52342068a31419c6c2d06ffa23425a7f76de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 12:56:21.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"52479562c49e294716477455de50af739aa449b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 14:54:29.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afaa3d0237834896b9262193f1099b92\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b6078d581994999b58768172f9dfb87 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a572704c3e334ce6934f16f3c69e182e : SUCCESS in 11m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db0a1da2e9df4e5eae3a2f8272f7c6fd : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec981a274a2949b286bf7549e618ad3d : SUCCESS in 12m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/7b523d516f2047b39fd959655c187dde : SUCCESS in 29m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8272c2d4a364ac09288d0ea4236fe95 : SUCCESS in 1h 16m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c9468cc789954e71a595718a1feac531 : SUCCESS in 1h 52m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b543bbc8c8274b37b56c04471d86b8db : SUCCESS in 36m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0849ada589e34f2f965ab1d5e2ae56e4 : SUCCESS in 17m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9e31fd48f5b84397a7dd20d3ab512c33 : SUCCESS in 1h 18m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6dca9113e727435da3238fe1c278f09f : SUCCESS in 1h 11m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a644e25ee584ff79b0f12915d2f119a : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":6},{"id":"d82343cbad30f3539da52deb82eee445c1dde2f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 14:54:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"ae8ef52da29a04d7ef1da80d1156354ca8f805d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 14:55:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e46b8a0886934ec085a18756de72783d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af3331af2b34adca133dd6adb9bfe57 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63a18f5d00fc48f08987ab3e3c1ce805 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":6},{"id":"e9d925ada0d23188ebcaee228a37b8fb8358fc43","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-18 07:50:31.000000000","message":"Patch Set 6:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_705/index.html : SUCCESS in 88m","accounts_in_message":[],"_revision_number":6},{"id":"c232f0a257e3e06ab431f666365e4386244d6e70","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-02-18 07:50:35.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/705/ : SUCCESS","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"56a944d52ccc0bc1debda83a96a26f21057f99b1","revisions":{"6f851c7205ada432941839823c1a0c3aac0befa8":{"kind":"REWORK","_number":1,"created":"2025-07-07 10:59:42.000000000","uploader":{"_account_id":37786,"name":"Pavithra Mahadev","display_name":"Pavithra Mahadev","email":"pavithra_mahadev@dell.com","username":"pavithra_mahadev"},"ref":"refs/changes/23/954223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/954223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/954223/1"}}},"commit":{"parents":[{"commit":"53245bce31549a29ce74dd49de6c9f7417868126","subject":"NFS driver: Fix fail to resize NFS volume with snapshots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53245bce31549a29ce74dd49de6c9f7417868126"}]}],"author":{"name":"pavithra_mahadev","email":"pavithra_mahadev@dell.com","date":"2025-07-07 10:27:07.000000000","tz":0},"committer":{"name":"pavithra_mahadev","email":"pavithra_mahadev@dell.com","date":"2025-07-07 10:59:04.000000000","tz":0},"subject":"Dell Powerflex: Enhance the logic to retrieve token for API communication","message":"Dell Powerflex: Enhance the logic to retrieve token for API\ncommunication\n\nImproved refresh token mechanism to retrieve token which helps to\ncommunicate with Storage device.The token will be retrieved every\n5mins. This helps to use the token for 25 mins, incase of\nPowerflex node failure, which takes approximately 5-6 mins to come\nup.\n\nImplements: blueprint dell-powerflex-token-refresh-update\nChange-Id: If5f74fb6bebb6762d1db471e814b94d2374ddfb6\nSigned-off-by: Pavithra Mahadev \u003cpavithra_mahadev@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f851c7205ada432941839823c1a0c3aac0befa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f851c7205ada432941839823c1a0c3aac0befa8"}]},"branch":"refs/heads/master"},"cdbe2dd297ceef9fcc71e6d109f87adf9e795ad9":{"kind":"REWORK","_number":2,"created":"2025-09-15 09:04:29.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/23/954223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/954223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/954223/2"}}},"commit":{"parents":[{"commit":"92c645f1f1e913b5b1cd8ad0227a251f03adec04","subject":"Merge \"Dell PowerFlex: Improve secret handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92c645f1f1e913b5b1cd8ad0227a251f03adec04"}]}],"author":{"name":"pavithra_mahadev","email":"pavithra_mahadev@dell.com","date":"2025-07-07 10:27:07.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-09-15 07:51:59.000000000","tz":0},"subject":"Dell PowerFlex: Update REST token refresh mechanism","message":"Dell PowerFlex: Update REST token refresh mechanism\n\nUpdate REST token refresh logic to retrieve a new token\nevery 5 minutes.\n\nDuring PowerFlex Management Platform (PFMP) node failure,\nno new authentication works. Keep using current token and\ntry to refresh every 1 minute until PFMP is back online.\n\nUse the new token and refresh token every 5 minutes.\n\nImplements: blueprint dell-powerflex-token-refresh-update\nCo-Authored-By: Yian Zong \u003cyian.zong@dell.com\u003e\nChange-Id: If5f74fb6bebb6762d1db471e814b94d2374ddfb6\nSigned-off-by: Pavithra Mahadev \u003cpavithra_mahadev@dell.com\u003e\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cdbe2dd297ceef9fcc71e6d109f87adf9e795ad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cdbe2dd297ceef9fcc71e6d109f87adf9e795ad9"}]},"branch":"refs/heads/master"},"274ca6513093e881edea008999962191067a0ed1":{"kind":"REWORK","_number":3,"created":"2025-09-15 13:28:51.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/23/954223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/954223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/954223/3"}}},"commit":{"parents":[{"commit":"92c645f1f1e913b5b1cd8ad0227a251f03adec04","subject":"Merge \"Dell PowerFlex: Improve secret handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92c645f1f1e913b5b1cd8ad0227a251f03adec04"}]}],"author":{"name":"pavithra_mahadev","email":"pavithra_mahadev@dell.com","date":"2025-07-07 10:27:07.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-09-15 13:25:40.000000000","tz":0},"subject":"Dell PowerFlex: Update REST token refresh mechanism","message":"Dell PowerFlex: Update REST token refresh mechanism\n\nUpdate REST token refresh logic to retrieve a new token\nevery 5 minutes.\n\nDuring PowerFlex Management Platform (PFMP) node failure,\nno new authentication works. Keep using current token and\ntry to refresh every 1 minute until PFMP is back online.\n\nUse the new token and refresh token every 5 minutes.\n\nCo-Authored-By: Yian Zong \u003cyian.zong@dell.com\u003e\n\nImplements: blueprint dell-powerflex-token-refresh-update\nChange-Id: If5f74fb6bebb6762d1db471e814b94d2374ddfb6\nSigned-off-by: Pavithra Mahadev \u003cpavithra_mahadev@dell.com\u003e\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/274ca6513093e881edea008999962191067a0ed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/274ca6513093e881edea008999962191067a0ed1"}]},"branch":"refs/heads/master"},"e539541cc37cf0f622213a94af1a83ccc5300c19":{"kind":"REWORK","_number":4,"created":"2025-10-14 04:06:39.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/23/954223/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/954223/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/954223/4"}}},"commit":{"parents":[{"commit":"e23e859dfd570ff1bd784ec62fc0258ca5681dee","subject":"Merge \"Show volume attachment host information for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23e859dfd570ff1bd784ec62fc0258ca5681dee"}]}],"author":{"name":"pavithra_mahadev","email":"pavithra_mahadev@dell.com","date":"2025-07-07 10:27:07.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-10-14 03:48:05.000000000","tz":0},"subject":"Dell PowerFlex: Update REST token refresh mechanism","message":"Dell PowerFlex: Update REST token refresh mechanism\n\nUpdate REST token refresh logic to retrieve a new token\nevery 5 minutes.\n\nDuring PowerFlex Management Platform (PFMP) node failure,\ntoken refresh fails. Keep using the current token and try\nto refresh every 1 minute until PFMP is back online. Then\nreset the timer to 5 minutes.\n\nCo-Authored-By: Yian Zong \u003cyian.zong@dell.com\u003e\n\nImplements: blueprint dell-powerflex-token-refresh-update\nChange-Id: If5f74fb6bebb6762d1db471e814b94d2374ddfb6\nSigned-off-by: Pavithra Mahadev \u003cpavithra_mahadev@dell.com\u003e\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e539541cc37cf0f622213a94af1a83ccc5300c19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e539541cc37cf0f622213a94af1a83ccc5300c19"}]},"branch":"refs/heads/master"},"1d13873e587f1f52e2e5b48a6ae73ff355a83f77":{"kind":"REWORK","_number":5,"created":"2025-10-26 09:51:33.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/23/954223/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/954223/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/954223/5"}}},"commit":{"parents":[{"commit":"e23e859dfd570ff1bd784ec62fc0258ca5681dee","subject":"Merge \"Show volume attachment host information for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23e859dfd570ff1bd784ec62fc0258ca5681dee"}]}],"author":{"name":"pavithra_mahadev","email":"pavithra_mahadev@dell.com","date":"2025-07-07 10:27:07.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-10-26 09:50:35.000000000","tz":0},"subject":"Dell PowerFlex: Update REST token refresh mechanism","message":"Dell PowerFlex: Update REST token refresh mechanism\n\nUpdate REST token refresh logic to retrieve a new token\nevery 5 minutes.\n\nDuring PowerFlex Management Platform (PFMP) node failure,\ntoken refresh fails. Keep using the current token and try\nto refresh every 1 minute until PFMP is back online. Then\nreset the timer to 5 minutes.\n\nCo-Authored-By: Yian Zong \u003cyian.zong@dell.com\u003e\n\nImplements: blueprint dell-powerflex-token-refresh-update\nChange-Id: If5f74fb6bebb6762d1db471e814b94d2374ddfb6\nSigned-off-by: Pavithra Mahadev \u003cpavithra_mahadev@dell.com\u003e\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d13873e587f1f52e2e5b48a6ae73ff355a83f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d13873e587f1f52e2e5b48a6ae73ff355a83f77"}]},"branch":"refs/heads/master"},"56a944d52ccc0bc1debda83a96a26f21057f99b1":{"kind":"REWORK","_number":6,"created":"2025-10-28 03:19:44.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/23/954223/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/954223/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/954223/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/954223/6"}}},"commit":{"parents":[{"commit":"e23e859dfd570ff1bd784ec62fc0258ca5681dee","subject":"Merge \"Show volume attachment host information for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23e859dfd570ff1bd784ec62fc0258ca5681dee"}]}],"author":{"name":"pavithra_mahadev","email":"pavithra_mahadev@dell.com","date":"2025-07-07 10:27:07.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2025-10-28 03:18:54.000000000","tz":0},"subject":"Dell PowerFlex: Update REST token refresh mechanism","message":"Dell PowerFlex: Update REST token refresh mechanism\n\nUpdate REST token refresh logic to retrieve a new token\nevery 5 minutes.\n\nDuring PowerFlex Management Platform (PFMP) node failure,\ntoken refresh fails. Keep using the current token and try\nto refresh every 1 minute until PFMP is back online. Then\nreset the timer to 5 minutes.\n\nCo-Authored-By: Yian Zong \u003cyian.zong@dell.com\u003e\n\nImplements: blueprint dell-powerflex-token-refresh-update\nChange-Id: If5f74fb6bebb6762d1db471e814b94d2374ddfb6\nSigned-off-by: Pavithra Mahadev \u003cpavithra_mahadev@dell.com\u003e\nSigned-off-by: Yian Zong \u003cyian.zong@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56a944d52ccc0bc1debda83a96a26f21057f99b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56a944d52ccc0bc1debda83a96a26f21057f99b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
