)]}'
{"id":"openstack%2Fcinder~954520","triplet_id":"openstack%2Fcinder~master~I9285e051743a2745e82198e2ee0534f823b6c538","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-08-23 13:50:45.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2025-08-20 14:54:07.000000000","reason":"\u003cGERRIT_ACCOUNT_21129\u003e replied on the change","reason_account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2025-08-21 14:26:08.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2025-08-23 13:50:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9285e051743a2745e82198e2ee0534f823b6c538","subject":"NetApp - Extended Consistency group support for NVMe/TCP driver","status":"MERGED","created":"2025-07-09 15:17:07.000000000","updated":"2025-08-23 13:56:04.000000000","submitted":"2025-08-23 13:50:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"954520","meta_rev_id":"ae72487ca2439a41b4f86711f6f416428ca55389","_number":954520,"virtual_id_number":954520,"owner":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-23 13:50:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2025-08-23 11:56:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2025-08-20 14:54:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2025-08-21 14:26:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2025-08-23 11:56:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":38290,"deleted":true}],"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-09 15:20:12.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-09 16:03:47.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-09 16:15:22.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-07-09 16:19:05.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-07-09 16:26:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-09 17:10:00.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-09 17:54:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-09 18:05:15.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-07-09 20:55:46.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-10 05:46:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"CC"},{"updated":"2025-07-10 05:46:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"CC"},{"updated":"2025-07-10 05:46:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"CC"},{"updated":"2025-07-10 05:46:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2025-07-10 05:46:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2025-07-10 05:46:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-07-10 05:46:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"CC"},{"updated":"2025-07-10 05:46:07.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2025-07-16 15:41:14.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-07-16 16:51:40.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-07-16 19:48:32.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2025-07-18 16:11:24.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-08-19 17:37:18.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-19 17:38:01.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-08-19 21:04:49.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-08-21 18:58:18.000000000","updated_by":{"_account_id":38290,"deleted":true},"reviewer":{"_account_id":38290,"deleted":true},"state":"CC"},{"updated":"2025-08-23 11:56:41.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-08-23 11:56:59.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"ce575dbcb10538371613f745b3de4f2cb7f81eb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-07-09 15:17:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38beba8a5e340b0b28856d3cae917e7042d0524f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-09 15:20:12.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7af14c927e8a6a96723df32fb7347792562de718","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-09 16:03:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/954520 : SUCCESS in 44m 58s","accounts_in_message":[],"_revision_number":1},{"id":"567f32dec08abf96890c5c8f6e4c75635c9acb5c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-09 16:15:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1202/.","accounts_in_message":[],"_revision_number":1},{"id":"54c6517d452ce879d653357c5984bf70f803754b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-09 16:16:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1332/.","accounts_in_message":[],"_revision_number":1},{"id":"422e241618c5a4a22a31ed935bcc7d4fa5fbb3f5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-09 16:19:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/954520 : SUCCESS in 1h 00m 29s","accounts_in_message":[],"_revision_number":1},{"id":"22a6718f6ede8ef9f8ecad672f7314a0eb947b20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:26:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39873b521f3349a79ad97e03fc8894a5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01cef89a8f95460192096dfee811369c : NODE_FAILURE Node(set) request b79e40601e1f4c07b88db927d0d993b8 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/77d41dd8c6df415e9c9b9c91e6d36c98 : SUCCESS in 29m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6863b7ec75b447fbb72b2b6105a18e6c : SUCCESS in 30m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebdff60b318ff916b1b3b55e5c128f4982f9f173","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-09 17:10:00.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a275f5d4cdd2a1ec250c1b06fe324f0e89dd79cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 17:54:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/667deec05b204174bb1b479f724aac8d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94719b64e20c4c61b6f995c9cc00c8b5 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ea4f17fd7934c2fb74568b14a1e93a8 : SUCCESS in 7m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d47177922c5242beaac32a481873b343 : SUCCESS in 7m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97d108927c5a48588df1614ab2d2d6cb : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11a4ca5ef4c2413180e1034295bf4e78 : SUCCESS in 6m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/fa10315ac1154b579885757750973c56 : SUCCESS in 32m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9be88a6d81eb41eeaad11d906d08188d : SUCCESS in 35m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b78a8a614e94ab2bb6c80d2f5eea5ab : SUCCESS in 1h 20m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d951f9902659425aac3cd3a369c6a025 : SUCCESS in 32m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54a1d0e6b2f34da3b0aa774fbaa7c6e3 : SUCCESS in 10m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/24c49b959d1244379872530634bffe04 : SUCCESS in 16m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f26575fecea04d25b176a718ecf10dff : SUCCESS in 2m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7e76105b0bfd4168bbd983a0bf2394eb : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c4567551a2e442ec94960f22c37494b3 : SUCCESS in 7m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4fef4f0f1c7b4a85a410519bb4a1b5c3 : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ab015277e08e4eeeb1ff3a9ef3c9ebbe : FAILURE in 18m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9719a97ec37d4ceab6f64f4d5abd611c : SUCCESS in 3m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3d608b782fe04c52b6767d37ae567451 : SUCCESS in 1h 06m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/337df7d3f15d4a7a9205e2c44d91cccc : SUCCESS in 2h 22m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8eea16e9583d48258298bc283d1f7c79 : SUCCESS in 1h 46m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8da2e678c2c94ad79308ba9869f6110c : FAILURE in 22m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b9cdd8ae95464550ac343d1c537d13ac : SUCCESS in 29m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4f89c105f62c4d0a9a4e474f6778ec79 : SUCCESS in 35m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b86242f2600f493780ebc6bbd432c1f6 : SUCCESS in 1h 44m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/56820d54a2124f8e8482ead33d3553b0 : SUCCESS in 41m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b5b600cd5beb4695ad35115ddc19b22b : SUCCESS in 1h 27m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e66ab6900c1b45f6b338cd1974ba988f : FAILURE in 23m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2d1156c3db4d47abb6a641a504779b8f : SUCCESS in 1h 54m 26s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/717772b7b8a545cd9a737782dcc3f26d : SUCCESS in 1h 26m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f7afd12864a44708d6378962c6c4d29 : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2a05c74797f6cec08a3dbea46ebe21976d1be21d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-09 18:05:15.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0528f0ab754e4f088dbacf7c5efc4880\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e23724331c2b478ab08143859cc2ff48 : NODE_FAILURE Node request 200-0000072506 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3855529d95b34ac7aada640eba37b637 : NODE_FAILURE Node request 200-0000072507 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e5d729f9210f40b1bcdd1800341efac3 : RETRY_LIMIT Host unreachable in 50m 41s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7e9446a6f09b4fa891985e8f12927727 : RETRY_LIMIT in 25s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a38fb6a9e3c44469e0c3ca384072445 : RETRY_LIMIT Host unreachable in 46m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9c0e383869ea0ab315223593f135978e3e8e40d9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-09 20:55:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs-rest cinder-tempest-plugin-ontap-nfs-rest : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-iscsi-rest cinder-tempest-plugin-ontap-iscsi-rest : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"df576a0092a52096ec40602ff9e64f160152c8ad","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-16 15:41:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/20/954520/1/20250716190133/NEC-V-ISCSI : SUCCESS in 5h 27m 19s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/20/954520/1/20250716190133/NEC-V-FC : SUCCESS in 5h 39m 39s","accounts_in_message":[],"_revision_number":1},{"id":"592244014cbc4c1355f29425accc9fffac357954","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-07-16 16:51:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"042611da2e3bd221f2fa0a8161daade454f096fa","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-07-16 19:48:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d06a6e46b549a543b76eb4f9bb4f33d80e5f2d8b","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2025-07-18 16:11:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff777f8dd9a1e43e3df582588b5f67f91bb287bd","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-18 18:44:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbc71538576e79be7b3f41925f9dea44b35facbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-08-19 17:37:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1f590c37aff760030707ece493db696651ffdebe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-19 17:37:18.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/220/","accounts_in_message":[],"_revision_number":2},{"id":"7bb8fb07287b29803ef3dabb7c59440441571e46","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-19 17:38:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"4c695ae1658a28a700867e0d51273912b046a72a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-19 17:51:29.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/954520/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/954520/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"5d9afc7b9d56a5bde4e093dd95fc7f8b5528e343","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-08-19 18:04:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e763d389e10b46def79b770fbe09c1e90117d120","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 18:53:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02f37c9d4b954cbf9502c2d4183b0595\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2107d977c74445419aa426813a24f8c3 : SUCCESS in 26m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a629c6a2df944265b576bb746f78e4d6 : SUCCESS in 28m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9695206658cb4489b28690788ca79b1d : SUCCESS in 17m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6330f72faad4598da6cf542af05252b9ef956d62","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-19 20:43:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/172a503bab6c47298e57dc818da71cb5 : SUCCESS in 45m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2b5474355c304052bb900fc59f42e937 : FAILURE in 55m 08s","accounts_in_message":[],"_revision_number":2},{"id":"0c9d1d76c4ba1b3ab1b37df2a13429c990e59a65","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-19 21:04:49.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-954520-2/FC/ : UNSTABLE in 2m 20s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-954520-2/iSCSI/ : FAILURE in 3h 25m 35s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"0d259c35c3675db9dcfe12d8030de022f5e0a62f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-19 21:15:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/954520 : FAILURE in 1h 25m 39s","accounts_in_message":[],"_revision_number":2},{"id":"cc2cf2c32e097221e5a1d1e2f40664c515f3325d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 22:55:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d92c23872e44dca89db8fc221d3749b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8e2df6a8f448bbab15d12014ef7ac1 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19954b42382841608457a6e5198e9cd8 : SUCCESS in 12m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/726f557f859f4392920aa5ac6fd04046 : SUCCESS in 10m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26fe8b77b1c043118c74ad3399ae7a4c : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d00d5490de47c991d5846927a72570 : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/a5babdad65e44f699d3520c588390248 : SUCCESS in 51m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/60310fc26af942c69ab68e5e86fc62ae : SUCCESS in 52m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dca5cbb4bf0e4710969b23fe52290b1e : SUCCESS in 1h 18m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e43d6ab14c684b69af9994d6ab46ec67 : SUCCESS in 50m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cfe0c4691354167adef989c20814b1c : SUCCESS in 14m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd48fa0250c64548be6bab4fcc27b8ec : SUCCESS in 20m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5e023c0978f6493ea7f9217f4c83ece6 : SUCCESS in 3m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/884cdcef9167429a9528fec9f5c553ed : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b9c289854f1047c898171e573d0260b7 : SUCCESS in 7m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/642dcc87713e4f3f8a88c5693e4e4569 : SUCCESS in 10m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f4a063555b4b4d80a8e5ccd3b36224f2 : FAILURE in 17m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/617e4c34c081487b9e9a4f65b3fa71c9 : SUCCESS in 4m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f9366c1e277941449547a51a00eca745 : SUCCESS in 1h 36m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6dcd131750644382a3755897c0aaf171 : FAILURE in 2h 30m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d21355990bd24eadb8bfbc3cdf72dda8 : SUCCESS in 1h 49m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6f13d6c8bba04d608a0720dd94e91985 : FAILURE in 18m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/972769a0d82945ba96e6637e305c77bf : SUCCESS in 29m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1bccc6db4e964d08ac9613955725ecf2 : SUCCESS in 1h 01m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cda24f56d6bf4bfa895fdb6aaa13518c : SUCCESS in 1h 38m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f7dbeab03ece463aa3f03fbceb14845a : SUCCESS in 33m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aba3ec8d473f478dbc26e7c8e0f8ea57 : SUCCESS in 50m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/be869567923d468d87580e3b330d5a0f : FAILURE in 14m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4140ed2367c8442eaff896a9fe792716 : SUCCESS in 1h 22m 11s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07d828ab26854c9698a6a63ecc338678 : SUCCESS in 1h 16m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4712061d6b934a10bde22b1914e871e7 : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":2},{"id":"6a83c1c40096590d0cd9f0fc8c056dc08dd40aa0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-20 00:44:54.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1488/.","accounts_in_message":[],"_revision_number":2},{"id":"f6f365625f90028ad1fcc1704b11d9e125e912e4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-20 00:51:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1623/.","accounts_in_message":[],"_revision_number":2},{"id":"3cd0448bfdc936fc4633573ff35b672147671554","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-20 14:54:07.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b269c224a27bfffe7b7a9df34436c6bf7da56515","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2025-08-21 14:26:08.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"62765ffe7e8970f51c0f42ba8030ce35352637fe","author":{"_account_id":38290,"deleted":true},"date":"2025-08-21 18:58:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23c216a8f552f1c6ab9fb31b9d6d3e5ede2c9370","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-08-21 23:45:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/954520/2/upstream-check/cinder-tempest-plugin-ontap-fcp/a6239b4/ : FAILURE in 2h 44m 56s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/954520/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/6808e67/ : SUCCESS in 1h 03m 16s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/954520/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/973c6d1/ : SUCCESS in 1h 02m 16s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"143ef874cdceb5c0ca2bd6c8f9ff7b837e865b01","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-08-23 11:56:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"de4640354f857a3408fa931cf9a1653b83d29203","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-23 11:56:59.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0c9d667a8b6c9501cb02d127fb40be25ef0b5077","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 11:57:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"69eb91da7251f450b34ab09f0d218d4314f1436b","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-23 11:57:30.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/9dddf202c6fd4013bc03cf57017cb884\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/0368bbdfd6c541769802f2aeac308eff : FAILURE in 30s","accounts_in_message":[],"_revision_number":2},{"id":"b3b61378f898d338ca5f3a2cc65bb5071d399301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 13:50:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4372f1ae92644f76b82bf7d3e7c198f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f98d99a84a64414880ae4aa3d4fd43d : SUCCESS in 9m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b81ed9ee39cc40fcb510e787363e0d4b : SUCCESS in 9m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b230844473c44aeebbfca631750dda1e : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec6a34908ef94dcaa32c4d8acc3ea0ce : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/0ce0ff0a92df43ddad07931462b4a1bc : SUCCESS in 56m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e5df5e3dfd4942eb87cc46d828088f87 : SUCCESS in 1h 12m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63dfe545cb8e4e0aa92f0642192b3984 : SUCCESS in 1h 05m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83402a8474e3438f84742979182412f8 : SUCCESS in 1h 04m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/674f8913a46d446a98b7261760f315c8 : SUCCESS in 12m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6523ee280aea4f9888dc6409aafeb6b2 : SUCCESS in 1h 17m 57s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/186e70f3254748099ed788189415278d : SUCCESS in 1h 49m 08s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c74541f6595c4b6286ee42e6a2ed0f33 : SUCCESS in 1h 25m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0efc253c18634cd499b8f2c3cd18ff55 : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":2},{"id":"c99f072e6ee7954dea9a5b232b06e87e6baa7732","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 13:50:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ae72487ca2439a41b4f86711f6f416428ca55389","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 13:56:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ec8fba6dfb64fa4af67ee13edecc9c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1344f0ab73d46ee9e980f7696b36320 : SUCCESS in 5m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49c6c30576af44eb967a93b6dac34358 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2b44c385bc45e515987ccd6cbcbb652294ff131e","revisions":{"4610f5b4e06e21279048243d9f80309341afb897":{"kind":"REWORK","_number":1,"created":"2025-07-09 15:17:07.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/20/954520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/954520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/954520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/954520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/954520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/954520/1"}}},"commit":{"parents":[{"commit":"3380f01ff2876a6fdb6d4fe94984936277aefe90","subject":"Merge \"Hitachi: Fix to have compatibility of GAD volumes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3380f01ff2876a6fdb6d4fe94984936277aefe90"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-07-09 15:12:41.000000000","tz":-240},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-07-09 15:12:41.000000000","tz":-240},"subject":"NetApp - Extended Consistency group support for NVMe/TCP driver","message":"NetApp - Extended Consistency group support for NVMe/TCP driver\n\nNetApp already support the consistency group for NFS/iSCSI/FCP\nprotocol. Extend the same support for NVMe/TCP protocol.\n\nChange-Id: I9285e051743a2745e82198e2ee0534f823b6c538\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\nCloses-Bug: #2116261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4610f5b4e06e21279048243d9f80309341afb897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4610f5b4e06e21279048243d9f80309341afb897"}]},"branch":"refs/heads/master"},"2b44c385bc45e515987ccd6cbcbb652294ff131e":{"kind":"REWORK","_number":2,"created":"2025-08-19 17:37:11.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/20/954520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/954520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/954520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/954520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/954520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/954520/2"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-07-09 15:12:41.000000000","tz":-240},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-08-19 17:36:06.000000000","tz":-240},"subject":"NetApp - Extended Consistency group support for NVMe/TCP driver","message":"NetApp - Extended Consistency group support for NVMe/TCP driver\n\nNetApp already support the consistency group for NFS/iSCSI/FCP\nprotocol. Extend the same support for NVMe/TCP protocol.\n\nChange-Id: I9285e051743a2745e82198e2ee0534f823b6c538\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\nCloses-Bug: #2116261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b44c385bc45e515987ccd6cbcbb652294ff131e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b44c385bc45e515987ccd6cbcbb652294ff131e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
