)]}'
{"id":"openstack%2Fcinder~955390","triplet_id":"openstack%2Fcinder~master~I71b0ece4d98cf099a5487279ed28fdc0a4061c8b","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2025-12-13 02:23:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71b0ece4d98cf099a5487279ed28fdc0a4061c8b","subject":"Fix updating service_uuid","status":"MERGED","created":"2025-07-18 16:41:49.000000000","updated":"2025-12-14 11:15:36.000000000","submitted":"2025-12-13 02:23:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955390","meta_rev_id":"6c91fbbe5caab4a42b007b30c3d771f144e0241f","_number":955390,"virtual_id_number":955390,"owner":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-13 02:23:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-12-12 22:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-12-09 17:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-12-12 22:12:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"}],"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-18 17:43:34.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-07-18 19:20:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-18 23:54:54.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-19 00:12:27.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-07-19 00:21:57.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-19 00:30:56.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-07-19 02:24:03.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-07-19 16:36:39.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-24 14:16:03.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-25 00:53:43.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-25 00:54:02.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-09-04 18:19:51.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-12-09 18:22:14.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-12-12 22:12:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-12-13 00:34:34.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-12-14 11:15:36.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"69f845736b4c88c198d4c87b38a96f83787c70aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-07-18 16:41:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c130c340c8c158bb9d2d1533ebd83fa06a1f5d62","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-18 17:43:34.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3d6cf06e22754767a0ed181ab75574fb\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/91129aad435247219772c2c1ad585529 : NODE_FAILURE Node request 200-0000075537 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d59f109938754270b6bfcf4592c75786 : NODE_FAILURE Node request 200-0000075538 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b77c3c1cad6043ada7725557787505c1 : FAILURE in 1h 01m 14s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/af8048c313934cf3b9b28d5939f2d6e8 : FAILURE in 56m 58s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/713e1577e7c94a9f826fdb00f6f171e6 : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":1},{"id":"91535f12061c27f5774620ae7e8674acf23d8026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 19:20:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14a2a991136844c4a85ffee88f1e0cb2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b694217c1364677a6e70173f481b0e0 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba404208135e4a0f86d8e12f7c0ca281 : SUCCESS in 12m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1fd5aa31fcee4ac08612e39775e5a531 : SUCCESS in 9m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a6ffaaa5b3ee4247a6a11fd05a78ac83 : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f672809878d4f6b991f0a7a12f92bff : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/fac123e4878b42688e69a0b69b8decca : SUCCESS in 58m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8bf105dd161c4ecd938e7bf14a4e2259 : SUCCESS in 1h 01m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/02c1a34a436845aaa9d030ed0311593d : SUCCESS in 1h 53m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97d1cb0f0bed45c38920f1a7eb44d048 : SUCCESS in 1h 05m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0392482b5b3445b69894ee2a4c5c94b6 : TIMED_OUT in 40m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b4d2f1ca623e4fa1b6068c7327bb34cd : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dc15663a497e45749014c01c59f1232e : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ed812d5e2f394903947596ddb3e46bb2 : SUCCESS in 6m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/15d5edaba61947039370461c2972cf07 : SUCCESS in 8m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2227f3fe5e734545821f6a5f54229616 : FAILURE in 29m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e2c802845101402a835e7d0f16ffb943 : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7116d0958ca444ef985b1e886065eeb5 : SUCCESS in 1h 32m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/58de4a8153574e26b167a7c5e3d1b03a : FAILURE in 2h 30m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d0759aeb196b475d8c787990dfa89091 : SUCCESS in 1h 53m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f45a4eabf81444799769f3d11a5c6d03 : FAILURE in 19m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/598fe22aa6bd4f77b9c8658cf0545336 : SUCCESS in 27m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b2533552dd1348d8bcd05b1193c65ba4 : SUCCESS in 58m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/21c459a5c138438e95a77c6287dd5f43 : SUCCESS in 1h 19m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/302f25b244574b03a343905a1b0a481e : SUCCESS in 46m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7e6017ee34fe4ef39553095d79b55e4e : SUCCESS in 1h 10m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fa36519dfb7f4bf3808224464c368070 : FAILURE in 15m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9eb7de0811864c5dbdb1450c5ae170f4 : SUCCESS in 1h 36m 29s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64dbd3c2e0644f17b4c3da05c1077c37 : SUCCESS in 1h 20m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ab0a13526b342bd954052ef7aa1b177 : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1124ca6f827fe4a9c3cc71dcfbb9313adb50116b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 20:23:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c40b89977c4e41f3ad70f9b407905a3a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db5fa1413164415886aca85477e4e66e : SUCCESS in 17m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f0e50469b0254537874a6fee84bbb927 : SUCCESS in 24m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/64c39737d40143b2aa8269c8a8cb4698 : FAILURE in 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff41a4b1e8769cbbdae0f704db039614b78b973c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-18 23:54:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/955390 : SUCCESS in 45m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f746c17f2c278742c7c0ac0dfdbe36505d699cf5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-19 00:12:27.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-955390-1/FC/ : UNSTABLE in 3h 21m 54s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-955390-1/iSCSI/ : FAILURE in 5h 51m 41s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"bd59f1f9031d7a9e56c206dec3671232e11daeb3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-19 00:21:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/204c0cd3d4b84d0e8d8bca7fe1b1f508 : SUCCESS in 51m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7548f50bd021434b883e1a2781a73789 : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":1},{"id":"65323db254a08eb14cadbe6a64cc9ef5838f40e3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-19 00:30:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1411/.","accounts_in_message":[],"_revision_number":1},{"id":"5ba442ecfd3b1e74047bb75af642112f35dd1300","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-19 00:31:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1278/.","accounts_in_message":[],"_revision_number":1},{"id":"b43a4393ed74223d279f1c4ffce1c294f7567b92","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-19 02:24:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/955390 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8d7d0a184311bdb6adc178cb0e01cebe80196ee4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-19 16:36:39.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f7c7d361a53b4c0eb81c170ab358c998d56f17ba","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-24 14:16:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"9915771d64e4debbee71a3b403e2164f2b409e00","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-25 00:53:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/90/955390/1/20250725003345/NEC-FC : SUCCESS in 6h 16m 19s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/90/955390/1/20250725003345/NEC-ISCSI : SUCCESS in 9h 19m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b51e2a57d238bb0f7457cccdbf799c31c5d389cf","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-25 00:54:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/90/955390/1/20250725041007/NEC-V-FC : SUCCESS in 5h 29m 51s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/90/955390/1/20250725041007/NEC-V-ISCSI : SUCCESS in 5h 43m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3a043a76d18658242234b5a236cb6aafad8c9144","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-09-04 18:19:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2ad559542c46a406b41541e927a574816ec30fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-12-09 17:47:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b4d6d8410d908d24a620c4150337d9194f76118f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 18:10:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f59a089ecdd4e80ae21950495bff53f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32b5cf2a61cf4a01aeee375e10107a72 : SUCCESS in 13m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3cc13cc6381140389a1111cb4bd842c0 : SUCCESS in 15m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c65ac3016464b3491fd7dc182a9b7c4 : SUCCESS in 20m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f853540a8ad7bb18bf7fc4276a3de47462086320","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-09 18:22:14.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/158/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"34a6eeb5b41e79c024851c6646a199d2e2c59fd6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-09 18:44:39.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I71b0ece4d98cf099a5487279ed28fdc0a4061c8b .","accounts_in_message":[],"_revision_number":2},{"id":"e61226176af899cd5922c980529bb02152390ae2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-09 19:31:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2661/.","accounts_in_message":[],"_revision_number":2},{"id":"1fd91bd4881dc5ce4900f6e6633953e094a8ea36","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-09 19:34:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2808/.","accounts_in_message":[],"_revision_number":2},{"id":"09029c4cb80ba179138fadad0391ecee8ff4fad9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-09 19:34:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/152f60507b9543329e0e556750bb6a02 : SUCCESS in 46m 16s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9b1d47f1f0e1422689c4905f8c12da5b : SUCCESS in 50m 04s","accounts_in_message":[],"_revision_number":2},{"id":"828bf26d281e4b53341a3461dc7f1dcc3eea8b78","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-09 20:19:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/955390 : FAILURE in 1h 08m 40s","accounts_in_message":[],"_revision_number":2},{"id":"95b8b42523b6d8e8fdd2b5fded1e6a18d95d1582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 20:56:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3ffc3863bb04cfa87c538c9623a6987\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2f7c64a759248488285f95dfd8a7cfc : SUCCESS in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f99a74b9f9074c0f8090df66ab00c157 : SUCCESS in 11m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a9f0413840b4f6cbabef29b625a4bac : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3116011639f4d39a382e3c8bb49ced4 : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/2037d191e7274c1e946e3dfe704827d5 : SUCCESS in 1h 00m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/32114da416504ad4a7f28bcd631638e0 : SUCCESS in 59m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1239cc73f15475eae0a88b8e6a13df8 : SUCCESS in 1h 45m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4eebc3cb47bf46dd9c9497868d2192cd : SUCCESS in 1h 05m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d8efdc11ca34e3ebe6b2e0154364979 : SUCCESS in 18m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6550f0c69230459cacf7d0bb6563c17a : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f9565b2b5830416d90d7962abbfc949b : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0269dc01cdff415987713bf5ca320bbe : SUCCESS in 5m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2353775c4be0430dad0bc62b53518170 : SUCCESS in 8m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d5af6899184a4b52b3c934c48489b526 : SUCCESS in 31m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/03e2fbbe463c4347bea45b96545bbb63 : SUCCESS in 5m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/719c2e13c0b740fd9ad15618b9f8b0f4 : FAILURE in 1h 35m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b30811a1325e4a509d8260faddf77658 : FAILURE in 3h 02m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7f516dcec2bb4228945c60fce96b8a11 : SUCCESS in 1h 47m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/16531ba4241f44e687327374b6e4e118 : SUCCESS in 1h 58m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9c44f80a64954603a7ec25250ff103a3 : SUCCESS in 24m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/86affb7c8945468dbe4c7d26605cb203 : SUCCESS in 40m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9f7a021394c34c838afa718b6ba7a5c2 : SUCCESS in 1h 27m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c603a2046a414da8883848a9d66e2ee9 : SUCCESS in 43m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/675d0ad011b64f7f9d006567121d9d7d : SUCCESS in 1h 41m 14s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e660de12f4254030891b0ee6b171268d : SUCCESS in 1h 01m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/263f796c3e314019a4ec9a7ce99e9153 : SUCCESS in 1h 27m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fd6cd16506b4d1babc23633a9a8a284 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":2},{"id":"724cca9c7f1ddd332027b464c6b408c8d989b3c8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-12 22:12:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d125d221b65878ebfcf45aae08d3bc5081fe0dc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 22:35:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/932564fa85e2446cb45d9faef5b0d3f9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c334bfe7f904906861957682d95c9e0 : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/98dee725f14848988f0d6307e1a9432a : SUCCESS in 14m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9bc6a50f726c4947b56f551ad7f27b75 : SUCCESS in 17m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b9b1204ef24b5b73d2a8df066f2d91466cf22541","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-13 00:05:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8661f17d708d4fe5b1200714f562b4ce : SUCCESS in 1h 37m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6c6555b5f3ee4ba699dff09d28c5f717 : SUCCESS in 1h 48m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4355185aba37f0993cfd23d7c4270dab3f35e6b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 00:27:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/878936431ecc443a97f2cb66a6dec7fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba10edeafd2341a5be041c8a232ab713 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ecf4e3f71a841bd9409e82bcb80be64 : SUCCESS in 10m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5acfd4eaa02346c3a4416a58953c0f01 : SUCCESS in 14m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8222f776806b47518607ca1391271d7c : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/21ae78018341473f96d719f29231dcc6 : SUCCESS in 59m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9387d79eb0c8407a8ffe933fa91c7120 : SUCCESS in 59m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7e7a02ba2724097893eb038ef4e2f66 : SUCCESS in 1h 40m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a39b679c9e6c4ac68b957003c9df32b3 : SUCCESS in 59m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1a1759114a6d4573b51874ff761cd407 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b6fcb4126ccf482fa803529864aacbde : SUCCESS in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/838cf9ac4bd34085aeeb74973ebf460e : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/52bf6cdb1dbb4b60938218d4fc3ac4d0 : SUCCESS in 5m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0a971f8b73754530b7c1959e0ac0c0db : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/03ef9fa50a8444268b7b1589544e614d : FAILURE in 40m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ec57a43837e4430994c98508b3d19b28 : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/68beb1ae578d4b4f818734bbdc9ccf23 : SUCCESS in 2h 08m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4c78194577d04f00b487f8135f8aff99 : FAILURE in 22m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8156478f276a44f08cddb7fca6b69ff3 : SUCCESS in 1h 55m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/438f9f6b7a234c4597b234c124719e4d : SUCCESS in 2h 03m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/79decf4c0256479f8617823255873f71 : SUCCESS in 40m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/539b182bdbdd4babb60bc21b01625023 : SUCCESS in 37m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ff4e9e5189994fd48774e8408c92b15b : SUCCESS in 1h 49m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/77e91618b8e142bcb6d020fd2ef65354 : SUCCESS in 50m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/43f7f38db6164f41a6e91895ed2c23d0 : SUCCESS in 1h 12m 41s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/448d282b6c25405883574f3127e33923 : SUCCESS in 2h 09m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8e0d7210b2a4181b18af932292a9a75 : SUCCESS in 1h 47m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f0f76b1ac274836b5b91e484f8de572 : SUCCESS in 1h 08m 57s","accounts_in_message":[],"_revision_number":2},{"id":"dcbb5caaadcdb047b274c9421485fe7f6fed03e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 00:28:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"78e5e3d2fa095e625cc585ce00237eb7c4b6ef44","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-13 00:34:34.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_12_2025_00_27_57\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"a33e56f6a569880b878951393b8d7d9212fa0e53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 02:23:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3150f85199ef41308573eafab7bac571\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64cb53a94fcb453b888a3a81cb34de98 : SUCCESS in 7m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a478d0ccbba64a78ab3a7c481d247f4f : SUCCESS in 7m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3459a95092ea4d47b89421075c96a409 : SUCCESS in 15m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb4b6780972245c59a8bce779ea0c0f3 : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/a8a88f0cc7c24ec5959700e748067904 : SUCCESS in 59m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b5196528cfd74d22b652fe1f5af0f2de : SUCCESS in 58m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c49572adcf04a949acb0d94a6b8d63b : SUCCESS in 1h 23m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/179d5a0ecd5c4b8c8360f70c5465dc32 : SUCCESS in 56m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4f8fa14106024e4d97962552cbbdea87 : SUCCESS in 57m 32s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b0b8ee3bba754381beb9eaab55256905 : SUCCESS in 1h 51m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5067c46a9d594034b15f330f01f095c4 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":2},{"id":"aef4fcf410abafda75f0aadc77968d0bb163f665","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 02:23:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"02eaba9b11790499f8a2d56f6091ebc12659d2f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 02:26:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1d12f75537243e59c2bd852c9a36dcb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1bd7a0e72e4449783210c2c0075dd3f : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":2},{"id":"6c91fbbe5caab4a42b007b30c3d771f144e0241f","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-14 11:15:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-90-955390-2-fc : SUCCESS in 1h 6m 37s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-90-955390-2-iscsi : SUCCESS in 1h 9m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0fd54df6eed6298dd20ceb101ec11423422c2a1e","revisions":{"6c719f54c61dde191cb5b38cd1b7664423f884ca":{"kind":"REWORK","_number":1,"created":"2025-07-18 16:41:49.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/90/955390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/955390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/955390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/955390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/955390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/955390/1"}}},"commit":{"parents":[{"commit":"d6a14bbc7644e6b1b7c79d1e40ad0bdd41b4c810","subject":"Merge \"Fix flake8 warnings on flake8-import-order\u003e0.19.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6a14bbc7644e6b1b7c79d1e40ad0bdd41b4c810"}]}],"author":{"name":"Walter Boring","email":"waboring@hemna.com","date":"2025-07-18 16:35:06.000000000","tz":0},"committer":{"name":"Walter Boring","email":"waboring@hemna.com","date":"2025-07-18 16:36:54.000000000","tz":0},"subject":"Fix updating service_uuid","message":"Fix updating service_uuid\n\nThis patch ensures that after a successful driver assisted\nvolume migration, the volume\u0027s service_uuid is updated to\nreflect the possible destination service\u0027s uuid.\n\nCloses-Bug: #2117273\nChange-Id: I71b0ece4d98cf099a5487279ed28fdc0a4061c8b\nSigned-off-by: Walter Boring \u003cwaboring@hemna.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c719f54c61dde191cb5b38cd1b7664423f884ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c719f54c61dde191cb5b38cd1b7664423f884ca"}]},"branch":"refs/heads/master"},"0fd54df6eed6298dd20ceb101ec11423422c2a1e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-09 17:47:56.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/90/955390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/955390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/955390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/955390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/955390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/955390/2"}}},"commit":{"parents":[{"commit":"f38c2950b13a2cde2ae9f146cf525a9d54c8b830","subject":"Merge \"image_utils: Detect missing device before calling qemu-img convert\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f38c2950b13a2cde2ae9f146cf525a9d54c8b830"}]}],"author":{"name":"Walter Boring","email":"waboring@hemna.com","date":"2025-07-18 16:35:06.000000000","tz":0},"committer":{"name":"Walt","email":"waboring@hemna.com","date":"2025-12-09 17:47:56.000000000","tz":0},"subject":"Fix updating service_uuid","message":"Fix updating service_uuid\n\nThis patch ensures that after a successful driver assisted\nvolume migration, the volume\u0027s service_uuid is updated to\nreflect the possible destination service\u0027s uuid.\n\nCloses-Bug: #2117273\nChange-Id: I71b0ece4d98cf099a5487279ed28fdc0a4061c8b\nSigned-off-by: Walter Boring \u003cwaboring@hemna.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fd54df6eed6298dd20ceb101ec11423422c2a1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fd54df6eed6298dd20ceb101ec11423422c2a1e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
