)]}'
{"id":"openstack%2Fcinder~955731","triplet_id":"openstack%2Fcinder~master~I5ce07098fc97a68c4250922ee01d79239030812c","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-10-06 18:49:38.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-10-06 20:32:10.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-10-06 14:05:45.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I5ce07098fc97a68c4250922ee01d79239030812c","subject":"Run nfs-fips-py311 and lvm-fips-py311 jobs on master","status":"MERGED","created":"2025-07-23 20:48:23.000000000","updated":"2025-10-06 20:34:09.000000000","submitted":"2025-10-06 20:32:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955731","meta_rev_id":"e545ca30e7fbd00c8ef2fcc4efc02b3dfe074532","_number":955731,"virtual_id_number":955731,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-06 20:32:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-10-06 18:49:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-06 14:05:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-10-06 18:49:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-23 21:20:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-23 21:53:33.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-07-23 22:19:03.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-23 22:44:43.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-07-23 22:54:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-23 23:12:20.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-07-23 23:25:35.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-23 23:30:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-23 23:36:00.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-07-24 01:10:19.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-07-24 14:17:30.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-30 20:31:21.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-07-31 03:32:33.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-01 13:22:43.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-10-02 12:37:43.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-10-02 12:37:43.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"6bc783bca0d46ebb57ee689a3ebe56ee7d4d908d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-07-23 20:48:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87e21c4cd195f9ff732468e4c96e63cd775d7419","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 21:20:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c90bb26afe34ef7b69cb8bcd71044c2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9aebc7b5707a401bbdb05f7047bdb600 : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3baa6d12999487ea6df895c068fbe06 : SUCCESS in 24m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/201499d9df00421e9d19abf24afcc775 : SUCCESS in 28m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ff3a7144bf54c95abf1904c965795812083002d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-23 21:53:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1b02823d57da470a85ce3f2f1dc0836a\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2c750a0696304347b5654dca990d8565 : NODE_FAILURE Node request 200-0000076321 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8a980b660807479480706d327673945f : NODE_FAILURE Node request 200-0000076322 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c4cd0a1d736a4ea3ae8474df5e32f285 : SUCCESS in 1h 04m 21s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a85177a582004a6595ae25ab81babe05 : SUCCESS in 54m 26s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f7a75b6ff4d547b88ad25e4f10f21f9c : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":1},{"id":"007a8befeb3527fc095b847df9fb5552a7e057ba","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-23 22:19:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/041b4b2c85584d7da5ca9a5e14b3b7f8 : SUCCESS in 42m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9c062a2611d24ed380f00ae8b34c3dbf : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":1},{"id":"90f1092bc9bb01acbb10182637738a82c0038cd0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-23 22:44:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1309/.","accounts_in_message":[],"_revision_number":1},{"id":"b1fec1ec65defa78ad63a7b569d68a14559fed22","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-23 22:54:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/955731 : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":1},{"id":"af1944d192b3d3533f0b3387a66722ca4e7d846b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-23 23:12:20.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-955731-1/FC/ : FAILURE in 2h 23m 37s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-955731-1/iSCSI/ : FAILURE in 1h 49m 12s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"c86ee5d19b41310e8333e78c7773f34815e5b53f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-23 23:25:35.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b429f9bf487293a645c76414d45d899ac66866b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 23:30:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99911286b49e4a8694d7ad07636b9633\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec267f5203e4437bb5f4cdecbb81784c : SUCCESS in 7m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/076f08f128034370825f053cb534a2d7 : SUCCESS in 12m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71bdf6510dab425ea398f2471f2f9eb8 : SUCCESS in 11m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6356acc161724ac8a98853ca3471dcbf : SUCCESS in 17m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/189a3c66a18f461ab5b8977e5038aaa1 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/34f8636514c04d41b01cc95c43ea8ad9 : SUCCESS in 57m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/42f3c1eebe1f40d6b78d02a957ed25e4 : SUCCESS in 1h 04m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ee533e7934f4d928698efbd2980c1ba : SUCCESS in 1h 25m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7c2058c393d4d18a5dbc5d58f3b2198 : SUCCESS in 1h 03m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d552607789c942b5bb1d5dd7ce10f5c2 : SUCCESS in 18m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/738d50e878504ae0851dff4d910c75a0 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6f8e4c3be34841ada586234a3002e88a : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9248fbf22c4e4c3fa4c6fc2e88da0103 : SUCCESS in 5m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3e027488b9804e3ebea1370dc143f783 : SUCCESS in 10m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/af8fd54129e14ab48ee719161677ba2d : FAILURE in 30m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9de40335a2eb4324960275bcd028aec6 : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2d6dad3378e2431ca975ebb634292d31 : SUCCESS in 1h 49m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/124a4f8801a84e749f37292f442adccc : FAILURE in 2h 38m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/640ad08c3d894ab6a90e6b2b64e4e836 : SUCCESS in 1h 47m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/880dc1333b8a4964b56b6e89a107d7a0 : SUCCESS in 2h 13m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a488c6ae94a24f7ab3918de371196cd2 : SUCCESS in 36m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/89a5ce94c16a4fc693dcd14314305103 : SUCCESS in 1h 05m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fb05fa2a9c424ea1abf253339510c565 : SUCCESS in 1h 36m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f45b297214d044cc8ac14728680a1544 : SUCCESS in 40m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3741117598d743d4af5acc4bf2bea8a1 : SUCCESS in 1h 17m 43s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3d49e7a8009249f5847002296d6bfbac : SUCCESS in 1h 47m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70f2c781ab084b1e97e4030459309deb : SUCCESS in 1h 19m 30s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cd2e999c75648e8abe40ed2e16bcc8b : SUCCESS in 1h 29m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e4c6149428b420baddcd01210a83b2c : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":1},{"id":"40327f5b41ff60483e316896c4512f5d67eb29f4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-23 23:33:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1443/.","accounts_in_message":[],"_revision_number":1},{"id":"cb45899ff18b49beebb92597ba713cb0f3835e1d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-23 23:36:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_07_2025_23_30_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"dc29c86dc18b59f03df6a44464196959b1bbd283","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-24 01:10:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/955731 : FAILURE in 1h 23m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5a4be7207662c7c0e0c705d073dcea3609ac43ec","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-24 14:17:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d507815210883864fbfee1e4e55ac8ab611c29b5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-30 20:31:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/955731/1/20250730235504/NEC-V-ISCSI : SUCCESS in 5h 28m 07s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/955731/1/20250730235504/NEC-V-FC : SUCCESS in 5h 36m 13s","accounts_in_message":[],"_revision_number":1},{"id":"670e7d281d1a954e2b3cada6803360c9f7e5db97","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-31 03:32:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/955731/1/20250731061029/NEC-ISCSI : SUCCESS in 5h 53m 04s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/955731/1/20250731061029/NEC-FC : SUCCESS in 6h 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1857dadd386b3e4d2c26de82f83caeb9b36d6a8d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-01 13:22:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cffea81e296c5aaa11fed9664f3f2010d6726a54","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-01 13:24:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e5bd8cd53dd24b54acf03e5373800837 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0e24ab067f8443cea5369098836c33a1 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"86e434add6266e96dbc5e5ccaeeef0e3feb0dc89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 15:14:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b677426782ad48a6950474b0536c32b6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a986be24ec504f4ba2b393e2a26c12cf : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/83944fceaf4644a8af584a1736119232 : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/966025cda4e445b5ac9a01a6ae37a7ac : SUCCESS in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48f815daff254703929ca36007301268e809a016","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 16:01:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0003582dd404994a65d117497688fa1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff93aa68ce83435d9b580401387c268a : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e01328106bd4c91b95df4f514862a47 : SUCCESS in 7m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/702d3206864b43dc994f94fd2a8b39c6 : SUCCESS in 11m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3cc5bf6525264b418d8acf8aa678a31d : SUCCESS in 15m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb0140b8656448a801bd361cd0a38a6 : SUCCESS in 11m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/99d3c29562604bceb5f06ad2642ad099 : SUCCESS in 58m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0825a973fe894472a7a404ab85a1bd23 : SUCCESS in 1h 07m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18b735333bbf49b69104c1359aa8f623 : SUCCESS in 1h 38m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f057c371cd943b1a129117893d7fb47 : SUCCESS in 51m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7fda263fc27e40a086b86fa651581be8 : SUCCESS in 22m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5c32d5c55e614263a04a70e5fc1bf2dc : SUCCESS in 3m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ab647bfe488948dc80c4358f4cac946c : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2b0ef4e1951240e6adc65b5626677be5 : SUCCESS in 5m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/750eef3ae665439c9e07f720a84a74be : SUCCESS in 8m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce976c84144c48d59936b44c25677cba : FAILURE in 18m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/34d47f9df1ec4708b078126026e715db : SUCCESS in 3m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6606e3d754184be2aca8342af66eb193 : SUCCESS in 1h 53m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8dd7a81e278a416f88c37661ff56ee7d : FAILURE in 2h 32m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d5591c3bfcba4629909a232d16492c6f : SUCCESS in 1h 46m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d63cdd74b5724c2eaa39315164d8c5f1 : SUCCESS in 2h 03m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1ef4720c56c445b187cd5fba2579e488 : SUCCESS in 32m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/551a00fc762c4a79aa9535954efa2653 : SUCCESS in 34m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ffa01a608e614aafae6739c470220a8d : SUCCESS in 1h 25m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/05613b1ab2ef47b48e6d4fd2fec6031e : SUCCESS in 25m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/800c512a133c4ab0a2d1c3b9f589f719 : SUCCESS in 1h 31m 46s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/94a88ed1b11f49d290c2df90a9ada0a6 : SUCCESS in 1h 43m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4427ca2995c4539a6df2fa4f721deaa : SUCCESS in 1h 46m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8940e193c42649d68898a46e1253400f : SUCCESS in 46m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2b097791e9d4c29925ea1d3869e62e76dd521cd7","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-10-06 14:05:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a236de3c653142534656b5bab67a5b82ef55caf1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-06 18:49:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85b83736e5e17e319e060adba0e06aac7798e1d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 18:50:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"747b5831d64e0e911b1738e0c31911cb1bdb7f52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 20:32:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7fbcb8b553f4037a2b4b8e116adcfe4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8d727895435400a9ebe376921e2d14a : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11bfc5b42e00409fba495e5520133e01 : SUCCESS in 9m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41bce509914248a6b2066f9ef24090d1 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc8599095cbb40ce824f7e92b540d9b5 : SUCCESS in 11m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/6364b33d258d42f89194b6cd5b2e0a00 : SUCCESS in 1h 00m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bea28df1b74b46a1b837ba7fbc8e8bd3 : SUCCESS in 36m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8dd7c3e1937d4e8dba817b764517b8f7 : SUCCESS in 1h 30m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/815016abd853439282b84b71e51d09bb : SUCCESS in 1h 03m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6c3400ce774348538eda70c3532a2e4a : SUCCESS in 49m 54s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1f257873fd3f458bb0fb834e946e26f3 : SUCCESS in 1h 35m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a38da6ef34f24aac8d72b4e023e7771d : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3391c0d4af39b2fd2ab6a0563a6ba3ceac1e9cb5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 20:32:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e545ca30e7fbd00c8ef2fcc4efc02b3dfe074532","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 20:34:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f92fdee07ed41db89d28b665d47a651\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2c2859a7dbf4039a7200e1ab1298a82 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c10bc5142185c3d716142cb416aa762b790f6065","revisions":{"c10bc5142185c3d716142cb416aa762b790f6065":{"kind":"REWORK","_number":1,"created":"2025-07-23 20:48:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/31/955731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/955731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/955731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/955731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/955731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/955731/1"}}},"commit":{"parents":[{"commit":"f54c9a23cd332aed1d625980b0b19e4e3f41222c","subject":"Merge \"Fix volume clone across cluster members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f54c9a23cd332aed1d625980b0b19e4e3f41222c"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-23 20:46:37.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-23 20:48:17.000000000","tz":-240},"subject":"Run nfs-fips-py311 and lvm-fips-py311 jobs on master","message":"Run nfs-fips-py311 and lvm-fips-py311 jobs on master\n\nFor the master branch, use the version of these jobs\nthat runs on python 3.11.\n\nDepends-On: Ibef46e0ff43352fe92ebfc801193a298e1494acb\nDepends-On: I4f861f6ec5ed5ff71ef3d4c17e696c2b75c9e4c9\nChange-Id: I5ce07098fc97a68c4250922ee01d79239030812c\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c10bc5142185c3d716142cb416aa762b790f6065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c10bc5142185c3d716142cb416aa762b790f6065"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
