)]}'
{"id":"openstack%2Fcinder~955889","triplet_id":"openstack%2Fcinder~master~I21bd59e0e05c506c5f9f17dc84478f27999b7281","project":"openstack/cinder","branch":"master","topic":"shared-volume-cluster-fix-conflict-with-clone_across_pools","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-12-15 20:28:06.000000000","reason":"Change was abandoned"},"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2025-12-15 20:28:06.000000000","reason":"Change was abandoned"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-12-15 20:28:06.000000000","reason":"Change was abandoned"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2025-12-15 20:28:06.000000000","reason":"Change was abandoned"},"38151":{"account":{"_account_id":38151,"name":"Ayhan ilhan","display_name":"Ayhan İlhan","email":"ayhanilhan@outlook.com","username":"ayhanilhan"},"last_update":"2025-12-15 20:28:06.000000000","reason":"Change was abandoned"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2025-12-15 20:28:06.000000000","reason":"Change was abandoned"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-12-15 20:28:06.000000000","reason":"Change was abandoned"},"33197":{"account":{"_account_id":33197,"name":"Mesut Muhammet Şahin","email":"mmessah@gmail.com"},"last_update":"2025-12-15 20:28:06.000000000","reason":"Change was abandoned"},"36045":{"account":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"last_update":"2025-12-15 20:27:32.000000000","reason":"\u003cGERRIT_ACCOUNT_36045\u003e replied on the change","reason_account":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-01 08:45:11.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I21bd59e0e05c506c5f9f17dc84478f27999b7281","subject":"Fix volume clone across cluster members","status":"ABANDONED","created":"2025-07-25 11:42:48.000000000","updated":"2025-12-15 20:28:06.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e902afc3284989540ce4b657acfec728cffe4c04","_number":955889,"virtual_id_number":955889,"owner":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":38151,"name":"Ayhan ilhan","display_name":"Ayhan İlhan","email":"ayhanilhan@outlook.com","username":"ayhanilhan"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-15 13:31:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33197,"name":"Mesut Muhammet Şahin","email":"mmessah@gmail.com"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38151,"name":"Ayhan ilhan","display_name":"Ayhan İlhan","email":"ayhanilhan@outlook.com","username":"ayhanilhan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33197,"name":"Mesut Muhammet Şahin","email":"mmessah@gmail.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":38151,"name":"Ayhan ilhan","display_name":"Ayhan İlhan","email":"ayhanilhan@outlook.com","username":"ayhanilhan"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33197,"name":"Mesut Muhammet Şahin","email":"mmessah@gmail.com"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":38151,"name":"Ayhan ilhan","display_name":"Ayhan İlhan","email":"ayhanilhan@outlook.com","username":"ayhanilhan"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33197,"name":"Mesut Muhammet Şahin","email":"mmessah@gmail.com"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33197,"name":"Mesut Muhammet Şahin","email":"mmessah@gmail.com"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38151,"name":"Ayhan ilhan","display_name":"Ayhan İlhan","email":"ayhanilhan@outlook.com","username":"ayhanilhan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-25 12:32:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-25 12:53:36.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-07-25 12:54:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-25 12:58:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-07-25 12:59:46.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-25 13:28:27.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-07-25 13:30:47.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-25 14:14:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-25 14:19:55.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-07-28 10:37:16.000000000","updated_by":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"reviewer":{"_account_id":35507,"name":"Kiran Pawar","email":"kiranpawarpict2010@gmail.com","username":"kpawar"},"state":"CC"},{"updated":"2025-07-28 10:37:16.000000000","updated_by":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"reviewer":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"state":"CC"},{"updated":"2025-07-29 12:22:03.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-07-30 08:58:41.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-30 09:03:44.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-08-02 22:55:32.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-08-03 08:59:51.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-23 10:32:10.000000000","updated_by":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"reviewer":{"_account_id":38151,"name":"Ayhan ilhan","display_name":"Ayhan İlhan","email":"ayhanilhan@outlook.com","username":"ayhanilhan"},"state":"REVIEWER"},{"updated":"2025-09-23 10:32:10.000000000","updated_by":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"reviewer":{"_account_id":33197,"name":"Mesut Muhammet Şahin","email":"mmessah@gmail.com"},"state":"REVIEWER"},{"updated":"2025-09-23 10:32:10.000000000","updated_by":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-09-23 10:32:10.000000000","updated_by":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2025-09-23 10:32:10.000000000","updated_by":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-09-23 10:32:10.000000000","updated_by":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-09-23 10:32:10.000000000","updated_by":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-09-23 10:32:10.000000000","updated_by":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2025-12-12 21:36:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"ffa0d9330cb64f51403283c3692f134f67c71d3b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"date":"2025-07-25 11:42:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54a22cf09a17f74f7ec0612f807a2d6d90fc31bf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"date":"2025-07-25 11:59:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c74991e9fa1db70bac92715bd7c99c625f820de0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 12:32:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f99fc6fecd44585adcd3d38eca3d28e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/06bfb32241f1435492b76655ce55d870 : FAILURE in 18m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aace38e3f1234777b1560333dd0e26c4 : FAILURE in 21m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ab755688c47f4a14804069e656237971 : FAILURE in 30m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e2014b6c59cb54a6cddb9a786460e5408246053c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-25 12:53:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1318/.","accounts_in_message":[],"_revision_number":2},{"id":"7b588b4f1dad08b33948bf2dcda700075f1b64a2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-25 12:54:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6ddd284d27a94607938f25fe8b7e12d7 : SUCCESS in 46m 03s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/80c9e9273fa44804b17f4c37101fdfb5 : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":2},{"id":"5d6d3ea5c696b3cbf135f57bbe633beef7ea083a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-25 12:58:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b1a8b618031548ae9be270ea37ff4ea6\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/551de33037b14dba91bba6995c9efa06 : NODE_FAILURE Node request 200-0000076448 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2588e2e2dd0b4beba741006be6a97c56 : NODE_FAILURE Node request 200-0000076449 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8afa26df2d5348069230be20a80b967f : SUCCESS in 58m 12s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/29daee4519994e04827607eb10f01945 : SUCCESS in 47m 33s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b92d72e737445e5915438e7f54d3d21 : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":2},{"id":"8a2f6af0839cee4c46f134ca8fe02b67d08d5cb5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-25 12:59:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/955889 : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":2},{"id":"012b6b501ba59ca6d19baa80879ad9107b416ace","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-25 13:00:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1453/.","accounts_in_message":[],"_revision_number":2},{"id":"605daadee8028fe79fc2db566e6972900a0e894f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-25 13:28:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/955889 : FAILURE in 1h 24m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ed138d6c18ed253397cea895f7c86a4417fe0553","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-25 13:30:47.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"85a1022697c2da700bea40eb2a42e4a9e93b0fc9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-25 13:50:39.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"f7a5046d139d4b5832a18ec593683e927edcd416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 14:14:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af7e7ce920bc4fd298b1f8c4e2e0051d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/048e42d158dc40efae405892f2ea7483 : SUCCESS in 7m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1909e5992ead4054b9ee2421ea593f91 : FAILURE in 13m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d31c68e1ec4b4f008ffe55729389b5a8 : FAILURE in 10m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/160eafafdaaf4a9796d0c7f6f9108a0f : FAILURE in 13m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a32298050d21426e92adc6509ac7fbac : SUCCESS in 15m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/2b5c6f23f9894e8c8548bfdf042c462f : SUCCESS in 59m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9cf6ac450c845d78b8c803149383e44 : SUCCESS in 1h 00m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2c4782b85f443988cfef93782c83ed7 : SUCCESS in 1h 26m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10b437b3fbe04dd9a9486e5cfc18d3ce : SUCCESS in 59m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/636ae1155beb4d00b51c5e18eb001b29 : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c54a6fb29e84584a7662bd79d64f77a : SUCCESS in 7m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8014e34395aa464aa028bc774bed0019 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d1cde215dde74f71ace748f77d4f89f5 : SUCCESS in 6m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/11f793a37e37401c8031626a76fa29a3 : SUCCESS in 11m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a0570d39003e41d78db8b2d861e0ebdc : FAILURE in 36m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/192b9f8b656f4c898f36de31aa6f99ea : SUCCESS in 7m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/97bf8c715e2740ebb06550b468945083 : SUCCESS in 2h 01m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/618d825e0368474f9505c55ce385254b : SUCCESS in 1h 35m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d5f68febdd3e42c1937c8368aa87b450 : SUCCESS in 1h 45m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/54a06a449f9c41a192d77dcfb8ca8116 : FAILURE in 15m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c7d1f65cf74f4cac97f0bf5b80e0df82 : SUCCESS in 24m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9dacb6dab53d409cbd8abec2338224f9 : SUCCESS in 40m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0de96f7793064871b46c4a4ca64740fb : SUCCESS in 1h 52m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a5c07d265e034ab9811595a668557509 : SUCCESS in 30m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6414cdaa38ce4eb080d1eace1e1f09b9 : SUCCESS in 1h 39m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9a1ce5151b014286945a69b526a40836 : FAILURE in 24m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a0f57a003434728a4c169ab8042da87 : SUCCESS in 56m 17s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b49be6c102e540948cd49ed5bf42cf70 : SUCCESS in 1h 19m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50abb318b9f54ec7a5c5dd911f981a54 : SUCCESS in 43m 53s","accounts_in_message":[],"_revision_number":2},{"id":"e75aa0b8457c1fef03cddaf1582470d1a4864dc2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-25 14:19:55.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-955889-1/FC/ : FAILURE in 2h 36m 32s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-955889-1/iSCSI/ : FAILURE in 1h 41m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"628b895126b69c33bd8978f189aa12f6fabe9a07","author":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"date":"2025-07-28 10:37:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a0056d8168d825040004bd4fa2a632dbe4b178c9","author":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"date":"2025-07-28 11:40:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bdde7535966da728a3269856285d972e7700425b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 12:23:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ccb98c66709470095db3594822cd0f3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d521c02a46ef4c36a2b93c2a34556040 : FAILURE in 16m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32cbb15291264e98b67e02925de7553f : FAILURE in 23m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f376143d6b53454aa1b3f58fe2f29350 : FAILURE in 28m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e947445e8df497822419b41c7e8a16fd7821308a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 13:52:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/571f359dd20a46ba9cd457f974ee06d7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c0fa38a04e444b0ba0095032fc77090 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/106ec1440f4b496289f85ec21a958137 : FAILURE in 10m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f68f9b40c4eb4e48892f79691d158bc7 : FAILURE in 9m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4754ed580def46808d4d2e6c9febac04 : FAILURE in 12m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/625c38888fcc484382202ec96b1887c9 : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/a233591b2a0a4a6e8f246b119ac26274 : SUCCESS in 53m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bda8508b2876422d89b32965cfce977c : SUCCESS in 54m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/484187c01c634ee3ab14245b53599bba : SUCCESS in 1h 28m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e24b01500e44d1db13d301bf568ae6b : SUCCESS in 58m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/efd8924d18fe4034bd2a259901664899 : FAILURE in 23m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/95c375afb5994c269548db0195677d4e : SUCCESS in 4m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc52a22a486a49a89962e3194b0398b3 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9deabe10697146da8084e2c832cfcee3 : SUCCESS in 5m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fa0c1e36dbb64761b9e1616bb7bd6fa1 : SUCCESS in 8m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/69cb9af5539a4317abcc39c3dae0b61d : FAILURE in 31m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6063ec84ea3d478188fc6f4716d2d49c : SUCCESS in 3m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3736eb07b0084f30b6a65d0ee7d7ea50 : SUCCESS in 1h 41m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6a87bb60960448508b788095cb85f5a5 : FAILURE in 2h 11m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a55c946ee408438197c0c1fbcac29de5 : SUCCESS in 1h 54m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b760a7d6319c478d8214c3884be35ff9 : FAILURE in 16m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f90c11f7478b4f99a9ab0dd3e99e4d19 : SUCCESS in 29m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c1afc895fe0949f482228e2d60c4d3b1 : SUCCESS in 1h 02m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/caf9fe6090ad4fd0b0a84ffd61f071b8 : SUCCESS in 1h 21m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b03069d4749348e992a6a7a1a2206c37 : SUCCESS in 32m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/58b1cbcde8634fbabb9266d6e996db55 : SUCCESS in 1h 26m 27s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/906c1d9565d7437394bc20e15e1eb678 : FAILURE in 8m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a20bce221285498f915d700e4ff925a6 : SUCCESS in 50m 41s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/403edf7f182c49738246a002b57a1501 : SUCCESS in 1h 16m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6147b8bbcb4475aad40e24d32e1dafe : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4ac95693cc3e9ef05ec5e260d25e6a5da059d92f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-28 14:01:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2cd0958b84bf46c686c16921531b7ff2 : SUCCESS in 46m 17s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cba6a990a78f415586ebedf4b60de05d : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b6a80c1fd3e9062d579da1ee3104a6e7afe57795","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"date":"2025-07-29 12:21:54.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d2b54177382301c8d3a6ab42cfcfbee74837cd07","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-29 12:22:03.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/58/","accounts_in_message":[],"_revision_number":3},{"id":"2f252fa7bf0ccb4c63cbb4e15bd5cd66ca3939ca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-29 12:31:45.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/955889/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/955889/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"dec902c073865ca4438fd655c850787d74cafc2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 12:43:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/364d027a29c64eb496ad648fac6b67a0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/742f586fa7db4bed9dea56992177c9ca : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee93a3b29be24bca9f468e56ac811343 : SUCCESS in 17m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9cd19969a4c7430e927210d5d0f73ccb : SUCCESS in 19m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f6db975a7053ef7bc78e5561ddd39ec29cada153","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-29 13:11:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6b2bcdc75bad47d9be3225d48c14cb6f : SUCCESS in 45m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7895e0d8823946af84e39b239bd90818 : SUCCESS in 47m 05s","accounts_in_message":[],"_revision_number":3},{"id":"030ea9ea062d350548219025d4ef638b600aedf1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-29 13:18:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/955889 : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":3},{"id":"1d87be96bd907a3af456965bcccb66a6d6eca7a6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-29 13:25:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1464/.","accounts_in_message":[],"_revision_number":3},{"id":"600254fcef877a80c8e1779902dd2cd1572c0722","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-29 13:26:38.000000000","message":"Patch Set 3:\n\nChecking vars {item_url} {project} {patchset} {change}\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9d7468cead60488b927e00cdb06f3c72\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e5416c2ae961491ea7080cd5a3522ef9 : NODE_FAILURE Node request 200-0000076738 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d994361e27324b49b442ca8b821a1cee : NODE_FAILURE Node request 200-0000076739 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6e26e79cd57f4d3792522a4d679702aa : FAILURE in 54m 35s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/67e6d17eed6347b5a3374eccb44974f6 : FAILURE in 1h 04m 10s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cae58693941349a9a0e2a80d8c4c6813 : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":3},{"id":"2db8ec1f60fc11eab50cc539c6a74fef1de823c0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-29 13:28:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1329/.","accounts_in_message":[],"_revision_number":3},{"id":"ecf0c80649b2b831101b34a155adfd8ed752a00b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-29 13:46:36.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/955889 : FAILURE in 1h 23m 18s","accounts_in_message":[],"_revision_number":3},{"id":"f24640ffa39c318c5e7267f32df9b9f67d83f987","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-29 14:21:46.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"6e21f88f0c60c625409090bb6d4fdc2d1531628c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 15:13:48.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad65637a272c4f5da716d1372201d5fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a583c842de4228ae830978de38a8aa : FAILURE in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/814c263f82544d768b2a7ff300b17c8a : SUCCESS in 11m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/549eb50c926a417280a473f68690738f : SUCCESS in 11m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/962a55c4141e4128acb71da678c73b66 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/501b1f33901645468540704282256461 : SUCCESS in 13m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/c90625b7663f4d02b9bb8635780d79a3 : SUCCESS in 58m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9f8d57361834cfd9e02ec3ffd9f3ebb : SUCCESS in 1h 07m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ffe7098ab0e42209b61a641dabc221a : SUCCESS in 1h 34m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25d757db79294c4a8a030e413c48c2b5 : SUCCESS in 1h 01m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3674057ae828495cb0014e97d9ee0be2 : TIMED_OUT in 40m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/02b01a0e2d8d49129cebedf8da57f6df : SUCCESS in 4m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8086cc388d7841e6a8286ac6cf06bf6c : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/40113a729373474f86e7d91dd7f8e73a : SUCCESS in 5m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/87031f71f05d4f4eb7af395f2e147fb8 : SUCCESS in 10m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cdd50b8c5a604e58992eac23bc9b69d4 : FAILURE in 35m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f157844202554839ae58b180c38e4798 : SUCCESS in 5m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/59da2ecf1aa34de0bba6526fcfaf331c : SUCCESS in 1h 45m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b57c2d64190443a1b6810dbf383215c5 : FAILURE in 2h 36m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/27bbf5a219474a50afd48bcb25081796 : SUCCESS in 1h 26m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0fd536e79ec1402986ffdfc96f9fd7fa : FAILURE in 7m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a3c7bb7fad52421ca3a43b66cb04683e : SUCCESS in 32m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ae11b2562f694f639db41bdf6f0bdf8e : SUCCESS in 1h 06m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0b17df307688454ba60e6dd74be9f672 : SUCCESS in 1h 37m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4dce156d5e5a406387bb58cbe073c7fa : SUCCESS in 40m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a7d678ad36be4893b056da09ec12505b : SUCCESS in 1h 30m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9e8776e1c43d49a39c152acd1134969f : FAILURE in 17m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a665ab26b3af4deaa0238dc3aaba0db6 : SUCCESS in 1h 33m 20s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2493d32b57484bca9125aaf4a96f1382 : SUCCESS in 1h 22m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56349ffdf38b4412a75ebee928ad4818 : SUCCESS in 1h 06m 02s","accounts_in_message":[],"_revision_number":3},{"id":"9dc80babb65d6bc05460ac63701ffcaedc816a2f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"date":"2025-07-30 06:27:15.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"86db8cc6800f16e6bf7c6d57338db249fd81abfa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-30 06:27:26.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/60/","accounts_in_message":[],"_revision_number":4},{"id":"9ea4c4c201f345a370a1385f046f905944fd4a07","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-30 06:37:15.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/955889/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/955889/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"475152a9a3db0d64cd803b3cfff35bee4d0abc1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 06:52:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c8f3312a3094b7cbd8ca76ec21d15b5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b782628c89024add91dcea8f1bd5b312 : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/343a2775e4eb4ef092218dc4bcfa6a32 : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1a2909197874450ab3026870803de4d0 : SUCCESS in 23m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8f7c86c4358948f842fde3217ff58cde3c99f0a5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-30 07:26:15.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1d5e27289131473e8148723a5c0555ce\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9faf2ce6379a430eb062261aca2d0ab8 : NODE_FAILURE Node request 200-0000076801 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/368ac230962a437a96baef76c6af694b : NODE_FAILURE Node request 200-0000076802 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/76578ad7e0fe4488b77e2b02f8e70216 : SUCCESS in 51m 41s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0b86dfcf6ed24ef38f0a00bac4b9023b : SUCCESS in 52m 18s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de2bf1154c414e3eb50a35d6ff57909e : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":4},{"id":"ec8e2101553590e25316e88f2ddbbec4f18f7f8f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-30 07:31:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1331/.","accounts_in_message":[],"_revision_number":4},{"id":"ae24812144ed2cfa2d1127e7eedff6447485b7b8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-30 07:37:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1466/.","accounts_in_message":[],"_revision_number":4},{"id":"9aa6cdf2a459ba2a5f176eafdd852384dc80f44e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-30 07:41:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c4380c7658c0476f884e4a3b3d1be46a : SUCCESS in 1h 00m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/68e4b4e3190b4ef7b0c8d04a64f8dfa5 : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":4},{"id":"7d846dd61300784657031aab8005412e5a240b9b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-30 07:42:17.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/955889 : SUCCESS in 1h 13m 15s","accounts_in_message":[],"_revision_number":4},{"id":"8a3375beb11832c243433e26a5727d6253ad063c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-30 08:09:18.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/955889 : FAILURE in 1h 40m 35s","accounts_in_message":[],"_revision_number":4},{"id":"af192d3efc15089bbc4f7b1beec0ad3523bad3ea","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-30 08:16:33.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"3ffcabf6be1f803977615f992ed64ec81ee8d0c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 08:57:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08f00fb46491493cacff96b7fe05ded5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cad1271790c4b13b3df8c4c6bdadd23 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6fd6574136f4fbeb65c0192fb274efa : SUCCESS in 9m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d10905b3cb9d40f3b1b28fb2e1207af3 : SUCCESS in 11m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4d54b56e7b146d68b7346bcb25dfc58 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6167cf58430c474ea96922914260fffa : SUCCESS in 6m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/b8f655d34b544b5e9b9606765e9a5575 : SUCCESS in 44m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2464d236f4d4784baa1d5bd1ee47b1a : SUCCESS in 48m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/14a7d3900bd54614996121a7b120b535 : SUCCESS in 1h 19m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c29001ffe204b30bcacc65eb8668214 : SUCCESS in 44m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/33de43f79f064b3995c2ad2cc22f8579 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c2c3c451f39e40ef8bd0aa7095d71d96 : SUCCESS in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0bc6b79b2e1f4219a798f6f8760fa741 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/91b371695e784a0386735f477b93fc93 : SUCCESS in 3m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ace8c4f83ae34eecaaa8715eb02e5f07 : SUCCESS in 9m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/db8bc3233eea4280842106340c542fb6 : FAILURE in 17m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c8ed928bc8334444b32518f276e30815 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f9114f14492345a4b233139865f5635a : SUCCESS in 1h 45m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f6be5e4ff2ab46e7b9ca65da326337c1 : FAILURE in 2h 25m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/727e50de49424b9aa7236224708b7a6d : SUCCESS in 1h 43m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/eb082bca6208405992280311df445fbf : FAILURE in 18m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d6bff72e8afa4a69b4e74fd5dac6ead5 : SUCCESS in 35m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cb07a41546c942108bdd09b1533afb82 : SUCCESS in 36m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ecc7c19ef40a4b258e7f04886574c283 : SUCCESS in 1h 27m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/963dbf708b73428da89c2130ddb51ad3 : SUCCESS in 36m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ce827cfc44b345139a1db9e7ae1cd5c1 : SUCCESS in 49m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/eff9cf726b25433c95f0a6112aa4dc2c : FAILURE in 9m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89cdeb31efb94e9ba6d54f0cd1a39a0f : SUCCESS in 55m 44s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49f0cad98f8d452fb70f4876ea6f8954 : SUCCESS in 1h 22m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/623d5a20a0f746c3811f7bc89402f084 : SUCCESS in 45m 25s","accounts_in_message":[],"_revision_number":4},{"id":"99b56125399f94f8be76fa5e78451ab93d786c0d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-30 08:58:41.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"747f838029f6d6077bdbb161db6991481dee3477","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-30 09:03:44.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_07_2025_08_58_03\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"0bac864f3e5789b65f2128e53456809026d4f784","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-08-02 22:55:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/89/955889/4/20250803021906/NEC-V-FC : SUCCESS in 5h 25m 33s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/89/955889/4/20250803021906/NEC-V-ISCSI : SUCCESS in 5h 36m 23s","accounts_in_message":[],"_revision_number":4},{"id":"e851ab8154505558ef80f9e93c13e04fd215f997","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-08-03 08:59:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/89/955889/4/20250803113635/NEC-ISCSI : SUCCESS in 5h 56m 15s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/89/955889/4/20250803113635/NEC-FC : SUCCESS in 6h 23m 13s","accounts_in_message":[],"_revision_number":4},{"id":"32048e3bf18bbb19c9637f6f9a9171be71cd1bdb","author":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"date":"2025-09-23 10:32:10.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"9ade7c0eb18d0d8ccd276bf65736e3c51a4aa8d7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-12 21:36:49.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fb2c8eb38b59e879e1fcf2104c4d5d045df5387b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"date":"2025-12-15 13:31:46.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5c3f84ace2ffdffc6f00e79bf243c91423dd5ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 13:31:53.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 955889,5","accounts_in_message":[],"_revision_number":5},{"id":"0d37d163f95668e5398f5ee68be0e19b95b53408","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 13:31:54.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 955889,5","accounts_in_message":[],"_revision_number":5},{"id":"ad32c48280aeab66f5e8ca5f2b9ffd8cb4922402","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-15 13:32:05.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 955889,5","accounts_in_message":[],"_revision_number":5},{"id":"84b05c085a11a4a9a376163f134acf83fc968445","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-15 13:32:31.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"555eef18d1ad16244f8874746570e18f1f62f748","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-15 13:51:12.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2726/.","accounts_in_message":[],"_revision_number":5},{"id":"42342d2104269a09971faad689650ee45e38d84b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-15 13:51:17.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2873/.","accounts_in_message":[],"_revision_number":5},{"id":"cc240773dee6ba03cd549b2f1c6d484340348ca1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-15 14:31:42.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I21bd59e0e05c506c5f9f17dc84478f27999b7281 .","accounts_in_message":[],"_revision_number":5},{"id":"ef2398ad35f9568d152f016cbca2eb7e6c2d99e2","author":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"date":"2025-12-15 20:27:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e902afc3284989540ce4b657acfec728cffe4c04","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"date":"2025-12-15 20:28:06.000000000","message":"Abandoned\n\nAnother change fixed it already : https://review.opendev.org/c/openstack/cinder/+/958425","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d26274a92cca5a69b1e4898038351cfae534ab14","revisions":{"711c2565d149f2fec13d6d6623aed7d3b8600a0c":{"kind":"REWORK","_number":1,"created":"2025-07-25 11:42:48.000000000","uploader":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"ref":"refs/changes/89/955889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/955889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/955889/1"}}},"commit":{"parents":[{"commit":"f54c9a23cd332aed1d625980b0b19e4e3f41222c","subject":"Merge \"Fix volume clone across cluster members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f54c9a23cd332aed1d625980b0b19e4e3f41222c"}]}],"author":{"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com","date":"2025-07-25 11:42:48.000000000","tz":0},"committer":{"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com","date":"2025-07-25 11:42:48.000000000","tz":0},"subject":"Prevent conflict between commits below.","message":"Prevent conflict between commits below.\n\n- blueprint clone-across-pools\n  https://gitea14.opendev.org:3081/openstack/cinder/commit/f2427b461867e6c32a3e5573666ead748cd5e1b4\n\n- Fix volume clone across cluster members\n  https://review.opendev.org/c/openstack/cinder/+/935700\n\nChange-Id: I21bd59e0e05c506c5f9f17dc84478f27999b7281\nSigned-off-by: Yusuf Güngör \u003cyusufyusufyusuf@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/711c2565d149f2fec13d6d6623aed7d3b8600a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/711c2565d149f2fec13d6d6623aed7d3b8600a0c"}]},"branch":"refs/heads/master"},"49fac57e79b8b97db53e2b39e9d9c6986070b68b":{"kind":"REWORK","_number":2,"created":"2025-07-25 11:59:53.000000000","uploader":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"ref":"refs/changes/89/955889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/955889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/955889/2"}}},"commit":{"parents":[{"commit":"f54c9a23cd332aed1d625980b0b19e4e3f41222c","subject":"Merge \"Fix volume clone across cluster members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f54c9a23cd332aed1d625980b0b19e4e3f41222c"}]}],"author":{"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com","date":"2025-07-25 11:42:48.000000000","tz":0},"committer":{"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com","date":"2025-07-25 11:56:14.000000000","tz":0},"subject":"Prevent conflict between commits below.","message":"Prevent conflict between commits below.\n\n- blueprint clone-across-pools\n  https://gitea14.opendev.org:3081/openstack/cinder/commit/f2427b461867e6c32a3e5573666ead748cd5e1b4\n\n- Fix volume clone across cluster members\n  https://review.opendev.org/c/openstack/cinder/+/935700\n\nChange-Id: I21bd59e0e05c506c5f9f17dc84478f27999b7281\nSigned-off-by: Yusuf Güngör \u003cyusufyusufyusuf@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49fac57e79b8b97db53e2b39e9d9c6986070b68b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49fac57e79b8b97db53e2b39e9d9c6986070b68b"}]},"branch":"refs/heads/master"},"4aefb2ef524764497b2012dd2472e4992e5894b6":{"kind":"REWORK","_number":3,"created":"2025-07-29 12:21:54.000000000","uploader":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"ref":"refs/changes/89/955889/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/955889/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/955889/3"}}},"commit":{"parents":[{"commit":"f54c9a23cd332aed1d625980b0b19e4e3f41222c","subject":"Merge \"Fix volume clone across cluster members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f54c9a23cd332aed1d625980b0b19e4e3f41222c"}]}],"author":{"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com","date":"2025-07-25 11:42:48.000000000","tz":0},"committer":{"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com","date":"2025-07-29 12:20:28.000000000","tz":0},"subject":"Prevent conflict between commits below.","message":"Prevent conflict between commits below.\n\n- blueprint clone-across-pools\n  https://gitea14.opendev.org:3081/openstack/cinder/commit/f2427b461867e6c32a3e5573666ead748cd5e1b4\n\n- Fix volume clone across cluster members\n  https://review.opendev.org/c/openstack/cinder/+/935700\n\nChange-Id: I21bd59e0e05c506c5f9f17dc84478f27999b7281\nSigned-off-by: Yusuf Güngör \u003cyusufyusufyusuf@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4aefb2ef524764497b2012dd2472e4992e5894b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4aefb2ef524764497b2012dd2472e4992e5894b6"}]},"branch":"refs/heads/master"},"f1fe2b4835ee806a8790e917cf123c597a1c61ed":{"kind":"REWORK","_number":4,"created":"2025-07-30 06:27:15.000000000","uploader":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"ref":"refs/changes/89/955889/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/955889/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/955889/4"}}},"commit":{"parents":[{"commit":"f54c9a23cd332aed1d625980b0b19e4e3f41222c","subject":"Merge \"Fix volume clone across cluster members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f54c9a23cd332aed1d625980b0b19e4e3f41222c"}]}],"author":{"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com","date":"2025-07-25 11:42:48.000000000","tz":0},"committer":{"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com","date":"2025-07-30 06:26:55.000000000","tz":0},"subject":"Prevent conflict between commits below.","message":"Prevent conflict between commits below.\n\n- blueprint clone-across-pools\n  https://gitea14.opendev.org:3081/openstack/cinder/commit/f2427b461867e6c32a3e5573666ead748cd5e1b4\n\n- Fix volume clone across cluster members\n  https://review.opendev.org/c/openstack/cinder/+/935700\n\nChange-Id: I21bd59e0e05c506c5f9f17dc84478f27999b7281\nSigned-off-by: Yusuf Güngör \u003cyusufyusufyusuf@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1fe2b4835ee806a8790e917cf123c597a1c61ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1fe2b4835ee806a8790e917cf123c597a1c61ed"}]},"branch":"refs/heads/master"},"d26274a92cca5a69b1e4898038351cfae534ab14":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-12-15 13:31:46.000000000","uploader":{"_account_id":36045,"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com"},"ref":"refs/changes/89/955889/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/955889/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/955889/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/955889/5"}}},"commit":{"parents":[{"commit":"f54c9a23cd332aed1d625980b0b19e4e3f41222c","subject":"Merge \"Fix volume clone across cluster members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f54c9a23cd332aed1d625980b0b19e4e3f41222c"}]}],"author":{"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com","date":"2025-07-25 11:42:48.000000000","tz":0},"committer":{"name":"Yusuf Güngör","email":"yusufyusufyusuf@gmail.com","date":"2025-12-15 13:31:46.000000000","tz":0},"subject":"Fix volume clone across cluster members","message":"Fix volume clone across cluster members\n\nCurrently, when creating a volume from an image-volume (an image backed\nby Cinder), the code attempts to locate the source volume by filtering\nstrictly on the \u0027host\u0027 attribute.\n\nIn a clustered environment (e.g., Ceph with multiple volume nodes, or\nActive/Active configurations), the Cinder service processing the request\nmay have a different \u0027host\u0027 ID than the one assigned to the image-volume,\neven though they share the same backend storage cluster.\n\nBecause the host check fails, Cinder assumes the volume is not directly\naccessible and falls back to downloading the full image from Glance.\nThis negates the performance benefit of copy-on-write cloning and\nincreases network/storage load.\n\nThis patch updates the image-volume lookup logic to check for a matching\n\u0027cluster_name\u0027. If the current service and the image-volume belong to\nthe same cluster, the volume is considered accessible for cloning,\nregardless of the specific host processing the request.\n\nThis fix is necessary to support efficient image cloning in HA/Clustered\ndeployments and resolves potential logic gaps when used alongside the\n\u0027clone-across-pools\u0027 blueprint.\n\nReferences:\nBlueprint clone-across-pools implementation:\nhttps://gitea14.opendev.org:3081/openstack/cinder/commit/f2427b461867e6c32a3e5573666ead748cd5e1b4\n\nRelated Change:\nhttps://review.opendev.org/c/openstack/cinder/+/935700\n\nCloses-Bug: #1906286\n\nChange-Id: I21bd59e0e05c506c5f9f17dc84478f27999b7281\nSigned-off-by: Yusuf Güngör \u003cyusufyusufyusuf@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d26274a92cca5a69b1e4898038351cfae534ab14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d26274a92cca5a69b1e4898038351cfae534ab14"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
