)]}'
{"id":"openstack%2Fcinder~956168","triplet_id":"openstack%2Fcinder~stable%2F2025.1~Ib70219a9d085257c90a75ddcfcb935b4659fd28d","project":"openstack/cinder","branch":"stable/2025.1","topic":"new-location-api-support-stable/2025.1","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-10-23 06:15:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib70219a9d085257c90a75ddcfcb935b4659fd28d","subject":"Add support for glance new location APIs","status":"MERGED","created":"2025-07-30 13:36:12.000000000","updated":"2025-10-23 06:17:00.000000000","submitted":"2025-10-23 06:15:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956168-new-location-api-support-stable/2025.1","cherry_pick_of_change":909513,"cherry_pick_of_patch_set":7,"meta_rev_id":"9760507ae376be245b6865490dd7e8c7862f559d","_number":956168,"virtual_id_number":956168,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-23 06:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-23 04:39:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-22 12:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-10-23 04:39:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-30 13:36:37.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-07-30 14:04:32.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-30 14:14:31.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-30 14:15:16.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-07-30 14:42:30.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-07-30 15:25:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-30 16:21:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-30 16:22:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-07-30 16:27:33.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-07-30 16:50:50.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-10-22 12:55:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-10-23 04:39:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"e4e3fbb36fec9c2da237e5ff28391fd35d0412cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-07-30 13:36:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3bda7579e9aacf3e944618d2e063590945b5a68b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-30 13:36:37.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/64/","accounts_in_message":[],"_revision_number":1},{"id":"bd0ab7cab43835b9ef6a82ff2d47a997b8ee93f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-30 13:46:46.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/956168/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/956168/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"cda780133e726d52a8faf236757b61888c63be87","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-30 14:04:32.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f4a13bf29cadbf9d565737fb128af2e6b027bab6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-30 14:14:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/956168 : FAILURE in 36m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ac1af09036e6ccddbf841f39122dd88282796526","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-30 14:15:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/956168 : FAILURE in 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"bab2dda24e10a67512dd07de37c360af4cfe65af","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-30 14:42:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1470/.","accounts_in_message":[],"_revision_number":1},{"id":"5f2b1527ade8277d3366e134ead0cf5b76437fc2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-30 14:43:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1335/.","accounts_in_message":[],"_revision_number":1},{"id":"2d09b30f904ac9a2237776c8713fd77c5fc76257","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 15:25:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9e6450f244c4945b4c8d5624c45b90a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2eefc84d6e844ec6a8938dd77c188a1e : SUCCESS in 29m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/072ef0a2fa964c85b7125fcf36367174 : SUCCESS in 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0d181a00989d159cff434f16a14d4c9e7238ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 16:21:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20e346e491de40f69f8235052adcbef7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/717653584aaa4011a8873157bc6d7e0d : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b56a0abb10c4409bdb37e16dd825abd : SUCCESS in 12m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b83ac4d975504b2396c816cb0353f741 : SUCCESS in 12m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bd339d4d8fd48e59a969fd22c49b204 : SUCCESS in 7m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/945dcf9bd4df4aff9967b8372a6bcc1a : SUCCESS in 32m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78a8b2b691424b02a6540111def85588 : SUCCESS in 55m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/958d74a7bdbf4ba9b641b2215ee159e9 : SUCCESS in 1h 32m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f64f6ae98b8a40a59acfcd77b117bfaa : SUCCESS in 33m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4871b18dc38d442396eb0f06e2a4f1e7 : FAILURE in 26m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/87a35d77292a49569bc8a1386dff8019 : SUCCESS in 6m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6bad1aeb740b44708dfd7b2db9ad90af : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e086818c0e664b11822719563903479a : SUCCESS in 4m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/81cbf18d48804152a5e47442695be307 : SUCCESS in 5m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0a04e5cf0ad14390b2bdf8dc28f8242c : FAILURE in 21m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/815fc8b913e144b5820b30beb6c9ee51 : SUCCESS in 3m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/906b131caa8b4f95aab0fd56a7b77e4f : SUCCESS in 1h 02m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a2a6c6b9f0464d708714382f04d6889d : SUCCESS in 2h 34m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/083ab2fb21db432d85bcc675d4ac5ffa : SUCCESS in 1h 58m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0d715c5f123b434383eceba4a2ee053e : SUCCESS in 1h 51m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/56bc16452c124a849c5c081bea6ace54 : SUCCESS in 21m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cd3f2aa1368541dfbba5ea0a392ee2de : SUCCESS in 1h 07m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/54c887dc02014e2189c396d75b59ea38 : SUCCESS in 1h 33m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a3b2439984a446f688f97e67f796d13b : SUCCESS in 27m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8b184bbdc5e8468497becbe4e59da01d : SUCCESS in 52m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8fe0b7a91268419f90e617eced2709bd : SUCCESS in 50m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ca2a3e3a8f374aa98a756a7cdaf307ed : SUCCESS in 55m 45s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb94e524ced84594a5edf76bd4b8c6ae : SUCCESS in 1h 32m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00b2c4438a4641168895f718c8a65fbd : SUCCESS in 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"15cdd1cb1ccb8007098f40003beecd2a755c5c72","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-30 16:22:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"65be8d9d4eecd1e0f4ae504c83d4457e9691121d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-30 16:27:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_07_2025_16_21_56\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"67be1d15c922765324ee082b5ac11e7e1c2170f4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-30 16:50:50.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f52423db3ca8417392e47c172144026e\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/34ffea8ed50f44d9a2228f69e2b9a972 : NODE_FAILURE Node request 200-0000076854 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f45b8feceb0d44fb9539f1f56444ac34 : NODE_FAILURE Node request 200-0000076855 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bf90ef6697524862817ead0a46aaa477 : NODE_FAILURE Node request 200-0000076868 failed in 0s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f7cf214cfc3a4295937145d027101aaa : SUCCESS in 48m 26s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d01e2d47931848229e6b95eb32e87db7 : SUCCESS in 47m 12s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/461d5a2a24bc441c8c4906ae3111484d : FAILURE in 55m 11s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/01a25a1b83d548f6a2226d20720002fe : SUCCESS in 49m 04s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bbf4b9a990634469b757684b04d59626 : SUCCESS in 47m 45s","accounts_in_message":[],"_revision_number":1},{"id":"88ffe9b599bc845a4c6f5b297d59e5838dbc1172","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-22 12:55:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b67c0a09460aced71fd7ca59e85403791e9f85c2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-23 04:39:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d24bb6f9c8a2b10e3ec944b62b9a4936abe7c6a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 04:40:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2242f9a8571fff61cde64e5e5a7f635d07a72d8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 06:15:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35571c968d7b43bdbaa2c2f198e16a7f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc28dec72d0470387624ec9f93ee8b3 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f41fcca8ae246eb832b0d6c80b6840c : SUCCESS in 9m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e902ed98b8e34b65bd0ca7cc4e69ec3e : SUCCESS in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3004a3d60a4122bf9d744de1784977 : SUCCESS in 11m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/81104fc8f1704a2d87e868846d34ba0a : SUCCESS in 55m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4f4cf318d0d74e9abe3a480ad64f6766 : SUCCESS in 41m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2c0bd276d6f4bf8a14cef25d417e782 : SUCCESS in 1h 30m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14c16b198e324f82adbc9c36119c0200 : SUCCESS in 50m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5344b7d30acc4021b76981761f3f77fe : SUCCESS in 1h 03m 17s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6814bab6b404441aa14f3d16d45c6238 : SUCCESS in 1h 34m 09s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/238f9f2b0a2b48e0a2723795c27e32a9 : SUCCESS in 1h 07m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1a0e6cad6684c67992697b35857f62c : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ff06f63e490202b4e5d576f27e1d799dee07a5a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 06:15:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9760507ae376be245b6865490dd7e8c7862f559d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 06:17:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/977030117f114740b3cf7b936148ceca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9aa4502da3a4c18835bc6a8084f039d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ecb27b7554280374d38bacfbe55f0c7161050c18","revisions":{"ecb27b7554280374d38bacfbe55f0c7161050c18":{"kind":"REWORK","_number":1,"created":"2025-07-30 13:36:12.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/68/956168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/956168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/956168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/956168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/956168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/956168/1"}}},"commit":{"parents":[{"commit":"4649e4e55f5f6c56a0d411a082876bbd76cdd608","subject":"Merge \"HPE 3par: Ignore duplicate IP in iSCSI/vlan ip\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4649e4e55f5f6c56a0d411a082876bbd76cdd608"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-20 09:07:25.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-07-30 13:36:12.000000000","tz":0},"subject":"Add support for glance new location APIs","message":"Add support for glance new location APIs\n\nThis patch adds the call for ``add_image_location`` API\nwhich triggers the new location API workflow in glance\nthat addresses OSSN-0065. It is more secure and robust\ncompared to the old location workflow.\nThis call will be made when glance is using cinder as\na backend and we want to perform an optimized\n``upload volume to image`` operation.\n\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: Ib70219a9d085257c90a75ddcfcb935b4659fd28d\n(cherry picked from commit a39fa8fb2876a98b83570295504c0f5e46da7d6a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ecb27b7554280374d38bacfbe55f0c7161050c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ecb27b7554280374d38bacfbe55f0c7161050c18"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
