)]}'
{"id":"openstack%2Fcinder~957114","triplet_id":"openstack%2Fcinder~stable%2F2025.1~Ia1e35e1d0cc87791ef6cab6ac2bdd5a51f798aee","project":"openstack/cinder","branch":"stable/2025.1","attention_set":{"32537":{"account":{"_account_id":32537,"name":"ZhaoYixin","email":"zhaoyixin@inspur.com","username":"zhaoyixin"},"last_update":"2025-08-12 11:25:37.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia1e35e1d0cc87791ef6cab6ac2bdd5a51f798aee","subject":"Fix the incorrect size of the bootable volume","status":"NEW","created":"2025-08-12 11:15:28.000000000","updated":"2025-10-22 13:06:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"175eaf19bacf9cd7309e95f1f2e1c64fd831d626","_number":957114,"virtual_id_number":957114,"owner":{"_account_id":32537,"name":"ZhaoYixin","email":"zhaoyixin@inspur.com","username":"zhaoyixin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-12 14:07:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":-1,"date":"2025-08-13 17:18:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-12 11:15:38.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-12 11:25:37.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-12 12:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-12 13:11:48.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-12 13:28:05.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-12 14:05:49.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-12 14:07:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-12 14:17:01.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-08-12 19:59:53.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-08-13 17:18:34.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-10-22 13:06:58.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"}],"messages":[{"id":"f4df9182bdc99e660e24702791a0712f4dafa3d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32537,"name":"ZhaoYixin","email":"zhaoyixin@inspur.com","username":"zhaoyixin"},"date":"2025-08-12 11:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfc925f9a82b57ccba7fe3d7e13d4932f3c52fd7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-12 11:15:38.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2a35b0a25dbcd32d98b4789bd67e5b46b131ad9e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-12 11:25:37.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/154/","accounts_in_message":[],"_revision_number":1},{"id":"b5b6fe4c8ff0df341d57a2a7222b9501becba419","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-12 11:38:31.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957114/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957114/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"eca6da600583a9f537055f106fe7b991a9fa7b07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 12:21:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41b3d0388fc342f79018f753c7924194\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3426359436e446979c5b255fc498b277 : SUCCESS in 32m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4fa9767c14704cbcbcd20807cdf4b5a7 : SUCCESS in 26m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e38eda5baa04aebb4ad7881e541594607575d3b2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-12 13:11:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/957114 : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9e6ccba24b29fa919d39d4d889bf7ddee4e20282","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-12 13:28:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1415/.","accounts_in_message":[],"_revision_number":1},{"id":"e83ef0ec719b9a79df63bda153840279b393c9e7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-12 13:29:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1550/.","accounts_in_message":[],"_revision_number":1},{"id":"955408eedd4a80ab2d4e4bee3c23bb931739a5f9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-12 14:05:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/957114 : FAILURE in 1h 24m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b1c0e5ad0de9258aa1b911e26997f5578e6d36f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 14:07:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60026d3385674ace8c102ce311c98777\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/316545f539e04f19ac038e3d96ff287d : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24251d39fadb401fb8998edbe0cc55e9 : SUCCESS in 12m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ade87b0693140718ad08098a004dc9b : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c989f8770e047efa53e97c07618232e : SUCCESS in 14m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/68743e5262ad41f2b212c8f6f8e916ab : SUCCESS in 1h 02m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/36ae0c836b17497b9745887761534d2d : SUCCESS in 1h 00m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/131bbc54b51d4173a4789c722fd19049 : SUCCESS in 1h 31m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b9794e88d1045b0b6f0a9a7eb51cdc3 : SUCCESS in 1h 03m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce5198685579425dbef94f614eeafb21 : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9df69226df5c4224b62a54118aaadbcc : SUCCESS in 6m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/26e73e5b0f014fdb9e2791c967ba049a : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9d988b7dd19c4bc2a2cc49cedfeef63e : SUCCESS in 6m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d8761b8a91b948c8a8924b1007182804 : SUCCESS in 8m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0afe878fa5f74313a654890d72f127b0 : FAILURE in 35m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d2a2de3591ed4a57a3863b884f6cda09 : SUCCESS in 8m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5d8f1b9d90c34a269f8f40127ce34b65 : SUCCESS in 1h 50m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a90031e34330499ea6b199367074311a : SUCCESS in 2h 22m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/952b41a6fb3b407c805ed458e2f123ba : SUCCESS in 1h 51m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8671800a26f947cc9f13483a3cffea1e : SUCCESS in 1h 45m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4f519c0bb1374048b3c645185fdbe8c7 : SUCCESS in 36m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8dc19f5219804023a3ccc4982d59cd5a : SUCCESS in 1h 06m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/513672a7e1df48c692e400556ecce4ad : SUCCESS in 1h 37m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6c2604ffb2bc453d8d9f44bcd453a099 : SUCCESS in 44m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/77a6b59ab10a42f6b57a0dfd30d3c1ea : SUCCESS in 1h 57m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f1b161cfce28473093f385208a56c2e1 : SUCCESS in 1h 43m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/07ec5d76628d4e4688a9732bff1c0f43 : SUCCESS in 2h 09m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ea54c59f6f44080adbc48d32cac28ca : SUCCESS in 1h 29m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcd7085641b6444083616617bc966203 : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":1},{"id":"70e0416887dabf728c1f942b0af57d012a3d99c3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-12 14:17:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_08_2025_14_11_43\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"0da335bc720960def25f8761f34ff9a8f4f0c93f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-12 19:59:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/006849af96924cfabb2382fba26a49ea\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5a61aff5d7134ecf8e25dd68b6896eb2 : NODE_FAILURE Node request 200-0000078434 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/493f5056b3fd415887120dbc7f3f98eb : NODE_FAILURE Node request 200-0000078435 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7afc14eae1164931813f830f738539b5 : SUCCESS in 46m 33s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca1241a9ef274169a15311c636e697b6 : SUCCESS in 45m 14s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2ab90d7eb8914503b185b7c27475aa4e : SUCCESS in 44m 45s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fac6781c6eee4e4c954af1778f14febc : SUCCESS in 52m 25s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f3a1e15125eb47e6b2e8c779651407d2 : SUCCESS in 51m 26s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c14226447c674310a08b5f56646ab66f : SUCCESS in 50m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5b0d7a727658b94df84362e7f94185129ce6876d","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-08-13 17:18:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"175eaf19bacf9cd7309e95f1f2e1c64fd831d626","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-22 13:06:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bdebd9ca07212aed7a3dd1d84d733e1e821ccc0","revisions":{"2bdebd9ca07212aed7a3dd1d84d733e1e821ccc0":{"kind":"REWORK","_number":1,"created":"2025-08-12 11:15:28.000000000","uploader":{"_account_id":32537,"name":"ZhaoYixin","email":"zhaoyixin@inspur.com","username":"zhaoyixin"},"ref":"refs/changes/14/957114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/957114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/957114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/957114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/957114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/957114/1"}}},"commit":{"parents":[{"commit":"4649e4e55f5f6c56a0d411a082876bbd76cdd608","subject":"Merge \"HPE 3par: Ignore duplicate IP in iSCSI/vlan ip\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4649e4e55f5f6c56a0d411a082876bbd76cdd608"}]}],"author":{"name":"zhaoyixin","email":"zhaoyixin@inspur.com","date":"2025-08-12 10:40:49.000000000","tz":480},"committer":{"name":"zhaoyixin","email":"zhaoyixin@inspur.com","date":"2025-08-12 11:14:40.000000000","tz":480},"subject":"Fix the incorrect size of the bootable volume","message":"Fix the incorrect size of the bootable volume\n\nWhen enabling the image cache, after executing volume.update(module_update), set module_update to an empty dictionary to prevent incorrect volume size updates.\n\nhttps://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L1263-L1266\n\nCloses-Bug: #2120417\nChange-Id: Ia1e35e1d0cc87791ef6cab6ac2bdd5a51f798aee\nSigned-off-by: zhaoyixin \u003czhaoyixin@inspur.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bdebd9ca07212aed7a3dd1d84d733e1e821ccc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bdebd9ca07212aed7a3dd1d84d733e1e821ccc0"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
