)]}'
{"id":"openstack%2Fcinder~957115","triplet_id":"openstack%2Fcinder~master~Id2e6dd24f9eb32e11703ea5b884f7d8ed53568ee","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"Id2e6dd24f9eb32e11703ea5b884f7d8ed53568ee","subject":"Dummy change to test Netapp CI","status":"ABANDONED","created":"2025-08-12 11:19:37.000000000","updated":"2025-09-11 16:47:25.000000000","total_comment_count":12,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"8d39d5072a9ad4307e13db3ea1a4c148480a5b62","_number":957115,"virtual_id_number":957115,"owner":{"_account_id":38162,"deleted":true},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-12 17:47:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38290,"deleted":true},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38290,"deleted":true},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-08-12 11:19:48.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-12 11:21:11.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-12 11:38:33.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-12 12:36:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-12 14:11:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-12 14:13:20.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-12 14:28:18.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-12 17:03:17.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-08-12 17:53:25.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-08-12 22:53:16.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-13 17:00:55.000000000","updated_by":{"_account_id":38290,"deleted":true},"reviewer":{"_account_id":38290,"deleted":true},"state":"CC"},{"updated":"2025-09-11 13:51:03.000000000","updated_by":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"reviewer":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"state":"CC"}],"messages":[{"id":"2499d6372262ba6b839e676b355e33a4cf89b53f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38162,"deleted":true},"date":"2025-08-12 11:19:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd3385d2a63fa4c84840a04bc71ec868c1ec9a96","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-12 11:19:48.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"19f310c009de2792185ecb60daf7ef4075b30a8a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":38162,"deleted":true},"date":"2025-08-12 11:19:55.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"1662d081138f9999b5718d3bb04b99c23c6f7dfd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-12 11:21:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3bf8de0a418248fa935746757b778ef1 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9eef57d166e440a3929e8279e67f1455 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"5dadb1b8d4654ed2437180d853e3aa2fdd986d0f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-12 11:38:33.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/155/","accounts_in_message":[],"_revision_number":1},{"id":"0bbd65939c9cb11ba55cfc01b9987916e8be8f6e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-12 11:50:34.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957115/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957115/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"5d79217a3880a802f6369124944efb4b7a4456e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 12:36:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2068125cc7b545848de3b3cdf475e0db\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/196ac4aa33ea410e8479adfff0260113 : SUCCESS in 27m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/898e601b006747d2b65d6903a9b652c2 : SUCCESS in 27m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b546c3d0edb8482d8687693b13391ac2 : SUCCESS in 32m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e31ae29795c7e1a7d7c96317016994eac868e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 14:11:20.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/655f198e44a146be80e857a17a34e131\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86f5f8a135f54116852cceba2c5e7df4 : FAILURE in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4af9c33f59254e88b1073e9360f06d49 : SUCCESS in 13m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65611beab78e461d9acb950fce3aca86 : SUCCESS in 10m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2bbc9b84754a49a8b2fd51ae67427346 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07e517d8f514ddb9f0442e161f2c787 : SUCCESS in 14m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/72addd33c3ad4c90ae64783ea29210b9 : SUCCESS in 56m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2a361ef6f725449882b0fcd9ed8d9801 : SUCCESS in 1h 03m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f2c9b844c914dd98f4b12c8d9bbce6e : SUCCESS in 1h 31m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d01c13a6173a416f8dd7c3d5cf54e0f9 : SUCCESS in 55m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d1d7632bf856498a8503de30e0f4bc47 : SUCCESS in 15m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2718f204fbe74b18927cb5132ed5b719 : SUCCESS in 5m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a35f4328bc484994b6f915399ccdfcad : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4eb8f0a30b8e440d8160e80f776b0131 : SUCCESS in 4m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1c7147d64ce148ff849ddd12b0dd687b : SUCCESS in 8m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a4f4f30326d4494198ecfcdd97b3e89d : FAILURE in 24m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f8fbaa8506ff45149fbb396dc1aba492 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f980ead1de9940c58ddacf638b829aec : SUCCESS in 1h 55m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e698508de5204f448ccfb9b353e66946 : FAILURE in 2h 45m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/edf3f2f6ee594c599292abc1bb778711 : SUCCESS in 1h 57m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/226f5b11c40f4045a633df2a9dce56e6 : FAILURE in 15m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ec24512d82ed454ca59c60c199619748 : SUCCESS in 35m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2d7e1d1045ab4b5c8f124cb53fca416d : SUCCESS in 1h 03m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c5e0105b53d54217bacdb7f6f3f7100a : SUCCESS in 1h 42m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e81aac02167a44c28f02f415650410c7 : SUCCESS in 42m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6471a53cf0ca428d9f10e9118249d27c : SUCCESS in 1h 26m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/53fd66b52cb842b39da5ae6cd8d5a74f : FAILURE in 27m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4746b15562de44749b502a69a434751b : SUCCESS in 1h 51m 49s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c63e083ab6c4a9788b77872ea12f675 : SUCCESS in 1h 35m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7da350905a7a4f8b9903870f7a98edf8 : SUCCESS in 1h 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9bdf807ad32014f0c8c0464ceed21834c3683381","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-12 14:13:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/957115 : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":1},{"id":"df4a99c693099250f9ce1f519711565aab6a42ee","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-12 14:28:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1416/.","accounts_in_message":[],"_revision_number":1},{"id":"1c0b8da5f45ca53e76ea6b53101b6078617d6019","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-12 14:31:50.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1551/.","accounts_in_message":[],"_revision_number":1},{"id":"2d6c90818da6575863d9794d71a77593daefd644","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38162,"deleted":true},"date":"2025-08-12 14:49:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"629c631fdee8068c3410c09baee75c07b37b8a26","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-12 14:49:13.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"2f4f11c01487697dfaeaf005673a16549d0284a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-12 14:49:14.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/161/","accounts_in_message":[],"_revision_number":2},{"id":"dbc14e06d6ceee74c8e46b31c7c63244bb2fa0b3","author":{"_account_id":38162,"deleted":true},"date":"2025-08-12 14:49:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6b68f6d9a8058aec7dd172511e6ca4a6c5600c4f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-12 15:00:46.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957115/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957115/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"b46bd2ee7ea7c3c165300ca843792da4086e210e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-12 16:13:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/42303f442ee54bb4bdc41cd479540e5b : SUCCESS in 47m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/adf8e6a0b30848b0b8683e6ddc2a49ee : SUCCESS in 48m 08s","accounts_in_message":[],"_revision_number":2},{"id":"e8b056fa4b2deceb6ee6a14c477e40053dee4dfe","author":{"_account_id":38162,"deleted":true},"date":"2025-08-12 16:48:21.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5545bad1696342c066ce1be90680abd5695307c7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-08-12 17:03:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/56fc0bd/ : RETRY_LIMIT in 14m 13s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"cea039240817b2c41206865f744c79a894a767aa","author":{"_account_id":38162,"deleted":true},"date":"2025-08-12 17:08:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"15df2a821723d27ffe62f40e0cfa7a389620eb34","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-08-12 17:09:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp cinder-tempest-plugin-ontap-fcp : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"3459a8a8d84aad5885dbfd2ee38584dab6657998","author":{"_account_id":38162,"deleted":true},"date":"2025-08-12 17:22:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68ac251a809c6af2ec92d8106748e75fc80ad338","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 17:31:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82e636839fd242b9913872b5aa05fb0b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a7b71acd3a954741b3f1b66557c594f8 : SUCCESS in 24m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1495b3e0e4af4c11bbc533bf9396cc0c : SUCCESS in 28m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dbb2e76aef194f8ca2214bdcf436006f : SUCCESS in 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3fc81dece229d9157505bfbe39c1250bdaafde26","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-08-12 17:45:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-fcp/f24f8f0/ : FAILURE in 22m 11s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"a3b4b341e81cac6d0a39d30459911b82a0969133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 17:47:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a2a06b40364ee6b2ef06cddd19333f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a48a1d2ac2074adb8cdb61ca6198fb82 : SUCCESS in 8m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af662a6c60dd41b29744cb10ce1eed26 : SUCCESS in 11m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13cbcb1cb0e9470389827dd9b83df23e : SUCCESS in 11m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d6ae4bda59641f3b7f8b0220f3c925f : SUCCESS in 15m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0f015ba9df44aeb2cb866985ddef18 : SUCCESS in 14m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/f1fb2b870df84634ad535b931df37018 : SUCCESS in 1h 00m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d423b1da1aff40eab3ed2c078743ba9d : SUCCESS in 1h 01m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0d3ff4d493c24cd3b3b27654128fdd65 : SUCCESS in 1h 43m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c4657e3dff64aceb515d5090099b3af : SUCCESS in 1h 00m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bebc992937084c8695c40d68d3a855d1 : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/936a636462c24f57ba4a355e3809221c : SUCCESS in 6m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a607ca69df443f291cc2aecc1dbf568 : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9b70900faa494bac9ae5e0c5818fb1da : SUCCESS in 5m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c9f507f2e387468c96fb4ef25721485a : SUCCESS in 7m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2597835531714588b8538eab0f89c45b : FAILURE in 24m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/deb9c18de6d949aaa899c340a964d96b : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/87b440fee0c646f19f1c139ce1b8a816 : SUCCESS in 1h 53m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4a3e075b54ff42cc92a259734cfdad53 : FAILURE in 2h 52m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/55d190eaf7ce406584be5c0bb3e9b828 : SUCCESS in 2h 41m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4d06ec6951f14ec28f12a3ae7d41f15f : FAILURE in 15m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ccf2bcc589dc4eb6a7337b0675169f92 : SUCCESS in 43m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/575b7b1a82804177b9d745490c9b3404 : SUCCESS in 1h 04m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/60a01ee8575d424ebcf90c15a9fa35a4 : SUCCESS in 1h 38m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ab13fe7e39b0464c9c5e99428e3f2547 : SUCCESS in 44m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a97e7b5f06924d3d926a4678973340cc : SUCCESS in 1h 18m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8d68f7da1a1a46bcbb4126cfaa07f16b : FAILURE in 14m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b870574872a749f8bb2ffe18ea67063e : SUCCESS in 1h 47m 10s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67a31701a6f144d5bab9de9f7dc5108a : SUCCESS in 1h 41m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0717accc07c4914a2435be5d034c394 : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":2},{"id":"298146b17c09fa6b519c5c38d16deb668d8c5861","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-12 17:53:25.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_08_2025_17_48_05\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"d2c10c8a3c0f76a13c43e35a021653b908b36b7b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-12 20:08:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/957115 : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":2},{"id":"c8f418db45d343f6b39c469db573f38b4af322e8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-12 21:30:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1559/.","accounts_in_message":[],"_revision_number":2},{"id":"401a1ccaa05b9ac32c7b5cc490854a5c54a571fe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-12 22:01:09.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1424/.","accounts_in_message":[],"_revision_number":2},{"id":"c8ecd139ab800ab7d020a0f6cbf322df073dec67","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-12 22:53:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/957115 : FAILURE in 1h 25m 29s","accounts_in_message":[],"_revision_number":2},{"id":"0e6b70050e49c5ad011c78578d9619f151dfe73d","author":{"_account_id":38290,"deleted":true},"date":"2025-08-13 17:00:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22e42d18ba5cfd0c5b8c539da3432009338bfc3a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-08-13 19:14:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-fcp/d57cf7c/ : FAILURE in 2h 12m 58s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"5bb59c8ee5fc0d2ff2ece84fcf6caf92de1aa732","author":{"_account_id":38290,"deleted":true},"date":"2025-08-19 08:45:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b3bdfe1dec6c3dd13b366a89c65a80b5739ecd6c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-08-19 09:45:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/629d555/ : SUCCESS in 59m 41s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"da5544b35992dfe5624165c28072a1e55ae710fc","author":{"_account_id":38290,"deleted":true},"date":"2025-08-19 12:50:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"52e007aa9ce6b3e2edda8607280bd6431b6a9c3e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-08-19 15:24:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-fcp/c5e166b/ : FAILURE in 2h 32m 33s (non-voting)\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/08b0994/ : SUCCESS in 47m 42s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/195345d/ : SUCCESS in 56m 48s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f6cc79077be686de56a91fce3097fde6ebabb20a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-26 17:19:19.000000000","message":"Abandoned\n\nTaking up space in the review list; restore if you want to continue to use it.","accounts_in_message":[],"_revision_number":2},{"id":"1e265e147ffff1d063a608d2c5c599cc96e2bc6d","author":{"_account_id":38290,"deleted":true},"date":"2025-09-03 08:45:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b171e61a5ffcf8d5c6d8563adb58bed65af9daf","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-09-03 09:43:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-fcp/2e0e893/ : FAILURE in 28m 40s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/bf80183/ : SUCCESS in 57m 58s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/a993b5b/ : FAILURE in 22m 52s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"6c65dfa3baf716ce58f7a2ed6caf7fead84d971a","author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2025-09-11 13:51:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d39d5072a9ad4307e13db3ea1a4c148480a5b62","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-09-11 16:47:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-fcp/ff283c0/ : FAILURE in 2h 55m 35s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/4081e21/ : FAILURE in 1h 09m 23s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/957115/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/e10cc22/ : FAILURE in 1h 10m 20s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6ad1e30e8d89f56a79ececdbfbcee85bc7009d2d","revisions":{"f58ebac011e8e53b1674c46f8104f12d26cc2770":{"kind":"REWORK","_number":1,"created":"2025-08-12 11:19:37.000000000","uploader":{"_account_id":38162,"deleted":true},"ref":"refs/changes/15/957115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/957115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/957115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/957115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/957115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/957115/1"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Manideen","email":"manideen@netapp.com","date":"2025-08-12 11:02:49.000000000","tz":330},"committer":{"name":"Manideen","email":"manideen@netapp.com","date":"2025-08-12 11:19:17.000000000","tz":330},"subject":"Dummy change to test Netapp CI","message":"Dummy change to test Netapp CI\n\nChange-Id: Id2e6dd24f9eb32e11703ea5b884f7d8ed53568ee\nSigned-off-by: Manideen \u003cmanideen@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f58ebac011e8e53b1674c46f8104f12d26cc2770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f58ebac011e8e53b1674c46f8104f12d26cc2770"}]},"branch":"refs/heads/master"},"6ad1e30e8d89f56a79ececdbfbcee85bc7009d2d":{"kind":"REWORK","_number":2,"created":"2025-08-12 14:49:05.000000000","uploader":{"_account_id":38162,"deleted":true},"ref":"refs/changes/15/957115/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/957115/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/957115/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/957115/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/957115/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/957115/2"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Manideen","email":"manideen@netapp.com","date":"2025-08-12 11:02:49.000000000","tz":330},"committer":{"name":"Manideen","email":"manideen@netapp.com","date":"2025-08-12 14:47:11.000000000","tz":330},"subject":"Dummy change to test Netapp CI","message":"Dummy change to test Netapp CI\n\nChange-Id: Id2e6dd24f9eb32e11703ea5b884f7d8ed53568ee\nSigned-off-by: Manideen \u003cmanideen@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ad1e30e8d89f56a79ececdbfbcee85bc7009d2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ad1e30e8d89f56a79ececdbfbcee85bc7009d2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
