)]}'
{"id":"openstack%2Fcinder~957128","triplet_id":"openstack%2Fcinder~stable%2F2024.2~I582c27d8d8a8e22d03b2d68152cb467a6db606bc","project":"openstack/cinder","branch":"stable/2024.2","topic":"I582c27d8d8a8e22d03b2d68152cb467a6db606bc-stable/2025.1-stable/2024.2","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2025-08-19 14:22:29.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I582c27d8d8a8e22d03b2d68152cb467a6db606bc","subject":"HPE 3par: Ignore duplicate IP in iSCSI/vlan ip","status":"MERGED","created":"2025-08-12 12:45:04.000000000","updated":"2025-08-21 07:02:10.000000000","submitted":"2025-08-19 17:01:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957128-I582c27d8d8a8e22d03b2d68152cb467a6db606bc-stable/2025.1-stable/2024.2","cherry_pick_of_change":951818,"cherry_pick_of_patch_set":2,"meta_rev_id":"7f8ee881e6d9c04aeb62376e1758b7487cbb0d7d","_number":957128,"virtual_id_number":957128,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-19 17:01:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2025-08-13 17:18:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-19 13:26:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-19 13:26:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-12 12:45:28.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-12 12:45:28.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-12 14:56:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-12 15:01:47.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-08-12 17:05:00.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-12 18:24:18.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-12 18:30:09.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-13 17:18:00.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-08-19 13:26:29.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2025-08-19 13:26:48.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"6e82a2f5a82481f1dbeccfbc71608ce3e0f93c19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-08-12 12:45:04.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"6241b0cb2a862cd716303faeca5bfc2377558e97","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-12 12:45:28.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"21704631b323746291885475f1605bd0d2cddc19","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-12 12:45:28.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/158/","accounts_in_message":[],"_revision_number":1},{"id":"1c03275ed8d1b8c2a42d38b4601f9033a83544de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-12 12:56:45.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957128/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957128/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"e94e00b3bacd5d7c4ef2e035a681246f1db05671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 14:56:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f91ef8d2833c4044b7a5c2e5154837fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b667addc0c4248b18068cd56ed38d1cc : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98bd8e468f42464386dbf2fe20f5aca8 : SUCCESS in 12m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72b250a97da94919b23f697fd3508eab : SUCCESS in 6m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7c7e34ebf7b4c2c9378612cca36e12c : SUCCESS in 13m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc8645b590646b3bbf4a63ca24ef77f : SUCCESS in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/b58ec6f5f6bb4047a5d9f4c6d08d1d40 : SUCCESS in 58m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0cf1439df664b5bb59b893040283b77 : SUCCESS in 1h 17m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5206685b1ff24372bd68d2229793fa88 : SUCCESS in 55m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/137787a6f0fe4198a56f8a238d009ef9 : SUCCESS in 16m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/798989f1ebdd494a8c06d1383c607eae : SUCCESS in 15m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/92263aa88a1a4206942f2f90805005bf : SUCCESS in 5m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/04c7fd9a3357464aaa52decc17ba05f2 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95d302e14c804d809ba93dd177750bd4 : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a27b1942c45842d78c33a79df5167615 : SUCCESS in 8m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f6ad309f68644966914bb33892fd2807 : FAILURE in 33m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c75708b239e4134bd9c044a14314322 : SUCCESS in 5m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a31a54a461054b2a911eb8bbe237b473 : SUCCESS in 1h 56m 40s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/97d05d79067a49e1b840f573c0dc2d40 : SUCCESS in 1h 47m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4b214c10e19846bca292c5257f22ab89 : SUCCESS in 2h 00m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/40db4ef57bd5464b914aba6b1e58beda : SUCCESS in 15m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7d41b5bd9d334b1b901d6e7eae41adb3 : SUCCESS in 1h 03m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9e670574fde54bee89277e5d575625fd : FAILURE in 1h 11m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2eea4d82346b4ccdb3eed5afcc98aebe : SUCCESS in 42m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4149c2f1a25a44f1ad7e707e6ba5af81 : SUCCESS in 1h 19m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c86272c3bdb846ceb086aa3b6a12e3c0 : SUCCESS in 49m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/257b6dc5706949a48b368a6b53ca528c : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":1},{"id":"8f2e7358884c6d9527a1dfbb8f61ecd10fc38f0a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-12 15:01:47.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_08_2025_14_56_29\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f477747981774989cd2c60ac1b1bb848483b4a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 15:24:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d681cad74c2143bcb6004e6ae13e15be\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f6414aa5fc94e45bcdcb862ca93da94 : SUCCESS in 22m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fd67cc7823e94e0f8f8a0336bc8f4cfe : SUCCESS in 23m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19a7914fc0210dc2fce625fc7d6eb6b08f314b0a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-12 17:05:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/957128 : FAILURE in 36m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7c0af294da72386b1f6f637db44d6b8d6e2f5bb1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-12 18:24:18.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1556/.","accounts_in_message":[],"_revision_number":1},{"id":"3ea8b067e62c01caf72ce1ecc6a6b75c7c06b95a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-12 18:30:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/957128 : FAILURE in 35m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a975a4042d2cc620d074a3ea9745f7ffd67e0e4d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-12 18:53:08.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1421/.","accounts_in_message":[],"_revision_number":1},{"id":"589179dd247ae9ab28a0655c050d50cb06a20dce","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-08-13 17:18:00.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4623675030ca3ca4be69518af3ac88832f2f1ec0","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-08-14 04:44:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5003cfc7357342602b2470fc5737eb0ee03b8438","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2025-08-19 13:26:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29520d86e1c92c11ee606b5e6eef86d58fa67799","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-19 13:26:48.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ddf60ff0159d17064c2524a658d35bafb53233b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 13:26:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7165eaf494fcee6ff6bbcc1852cc5e5906cb2a4","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-08-19 14:22:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"320089fd2cb7d6eaa0e2e4747e8e11e7e05b0a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 17:01:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d77b0939b2ad40d5832c3dd0e261d041\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec04df15920c495182b3c814a2f111a5 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99805f6ab8454e9380239ad793a6a3d0 : SUCCESS in 11m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6de82a27c1c64deabc8efc55f42a264b : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6b9429a67c84686a57c3ba96aa046cd : SUCCESS in 6m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/2424f8d047af4d98b13a9e80a50c2c48 : SUCCESS in 45m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da065c7c38624a8c8753163bc5cc3e36 : SUCCESS in 1h 17m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28049593ff734d488cb512085eb814b3 : SUCCESS in 30m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7f63827f4c742c08753e2c59e90cbcb : SUCCESS in 18m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee0c60bfa64c47ecbf412569a7c44651 : SUCCESS in 1h 06m 52s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/55fb19833224471f9c2eeb4261f3917f : SUCCESS in 1h 55m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e029fd98a0f94540809ca1e7120b4e7f : SUCCESS in 26m 33s","accounts_in_message":[],"_revision_number":1},{"id":"09900a4f9f912df4783e4f2629778407a20ab867","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 17:01:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0bad44a6a4d902b34af49f8c5e5a265d7605dc42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 17:02:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19dddb787c254e219ce0352faec818c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255573a6bd274463bc257b55a2546e05 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08a2c82e348346d1b7e79e2509d20146 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1},{"id":"75a23ae12598e31bbf489b08dbef5f9d6c5fae7c","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-08-19 18:52:55.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttp://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/buildset/d90c4b5007c54649bb432a94b7c4913a\n\n- tempest-ci http://vast-zuul-ci-logs.s3-website-eu-west-1.amazonaws.com/t/opendev/build/45f58f64712d4ea987af85823fc04c89 : FAILURE in 5h 26m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7f8ee881e6d9c04aeb62376e1758b7487cbb0d7d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-08-21 07:02:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e96dcc4e9744833ccea2697c67198f55aba3ad22","revisions":{"e96dcc4e9744833ccea2697c67198f55aba3ad22":{"kind":"REWORK","_number":1,"created":"2025-08-12 12:45:04.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/28/957128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/957128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/957128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/957128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/957128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/957128/1"}}},"commit":{"parents":[{"commit":"77bfba4c2eea7703c8b4bd7a0d6a47b69ba318b7","subject":"[docs] Extra spaces breaking rst-\u003ehtml rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77bfba4c2eea7703c8b4bd7a0d6a47b69ba318b7"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2025-05-20 09:05:55.000000000","tz":0},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2025-08-12 12:45:04.000000000","tz":0},"subject":"HPE 3par: Ignore duplicate IP in iSCSI/vlan ip","message":"HPE 3par: Ignore duplicate IP in iSCSI/vlan ip\n\nAs seen in output of showport command below, there is possibility\nof same ip address (172.28.50.151) being used for trunk iSCSI ip\nand vlan ip.\n\ns0452 cli% showport -iscsi\nN:S:P State IPAddr         Netmask/PrefixLen Gateway TPGT . . .\n0:4:2 ready 172.28.50.151  255.255.0.0       0.0.0.0   42 . . .\n---------------------------------------------------------\n1\ns0452 cli%\ns0452 cli% showport -iscsivlans\nN:S:P     VLAN IPAddr        Netmask/PrefixLen Gateway . . .\n0:4:2 untagged 172.28.50.151 255.255.0.0       0.0.0.0 . . .\n-            5 172.28.50.240 255.255.0.0       0.0.0.0 . . .\n------------------------------------------------------\n2\ns0452 cli%\n\nThis patch checks for duplicate IP and ignores them;\nthus avoiding multiple calls (to create lun) for same IP.\n\nWhy this change is needed:\nBelow is the use case.\n\n1. initialize connection function is invoked.\n2. using trunk iSCSI ip 172.28.50.151, create lun invoked.\n3. With vlan ip 172.28.50.151, it again tries to create lun.\n4. And this fails because LUN is already created (with same ip)\nin step 2.\n\nExtract from cinder-volume log:\nDriver initialize connection failed (error: Conflict (HTTP 409)\n18 - LUN exists).\n\nSolution:\nWhile processing vlan ip, check if its same as iSCSI ip.\nIf so, skip the lun creation.\n\nCloses-Bug: #2112433\nDepends-On: I91cd5e262513b5427377ce1892e9acfe29e22b21\nChange-Id: I582c27d8d8a8e22d03b2d68152cb467a6db606bc\nSigned-off-by: raghavendrat \u003craghavendra-uddhav.tilay@hpe.com\u003e\n(cherry picked from commit 255ccd6c6cc6c64b154d9fea7fc9020e80bb4766)\n(cherry picked from commit df190f7f4c720a8fa920f0ef55536f1181d78951)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e96dcc4e9744833ccea2697c67198f55aba3ad22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e96dcc4e9744833ccea2697c67198f55aba3ad22"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
