)]}'
{"id":"openstack%2Fcinder~957190","triplet_id":"openstack%2Fcinder~master~Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60","project":"openstack/cinder","branch":"master","topic":"bug/2120509","attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-05-22 01:14:32.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"37798":{"account":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"last_update":"2026-06-19 19:49:39.000000000","reason":"Someone else replied on the change"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-05-29 03:00:19.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-05-22 01:14:32.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{"38702":{"account":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"last_update":"2026-06-19 19:59:43.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60","subject":"Add additional rbd options to improve download speed. For some rudimentary benchmarks, see: https://bugs.launchpad.net/cinder/+bug/2120509","status":"NEW","created":"2025-08-13 02:11:47.000000000","updated":"2026-07-12 10:39:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c67092e45e63a46b4008fb9d64bb001bbf7bf795","_number":957190,"virtual_id_number":957190,"owner":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-17 04:01:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-13 02:11:55.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-13 02:11:56.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-13 04:35:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-13 05:06:52.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-13 05:15:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-13 05:34:14.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-13 05:58:17.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-13 06:56:10.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-22 16:49:39.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 15:58:24.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-05-19 02:51:13.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-19 03:01:01.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-05-19 03:06:04.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-20 06:19:09.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-05-22 03:46:40.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-22 15:53:40.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-06-12 02:34:53.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-12 03:42:57.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-06-19 19:49:39.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"beeef909e067326eaedabe8f29919432ba67a6c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2025-08-13 02:11:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbf5ce8428db22e40c44b5a22bbab117aa7bfe3d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-13 02:11:55.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6025653cbf8e28970da758d4342dab6cee58d73a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-13 02:11:56.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/168/","accounts_in_message":[],"_revision_number":1},{"id":"99e980ab5472b9f57baaeed3947dc67cb62d866e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-13 02:24:22.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957190/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957190/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"b938432bfc9167e69f4191377cd1b11efc60afd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 04:35:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f780731e3f564916acec030d34eaf170\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b824b6889934240a034b3ed72294c97 : SUCCESS in 25m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cc7df7b9f9a246b4a0a74f697a9ed67c : SUCCESS in 33m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/29fcbce088a543d3b05389df2312c3e6 : SUCCESS in 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5fc1b2b64d29e202b6640ced67e14c90c818f82","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-13 05:06:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/957190 : SUCCESS in 56m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e0ab33bca3598f1825a9b89730b90817485dd9be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 05:15:13.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ba224b5b96a4e8c9e6fe3d2a50de10f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b7779a62fb46229f862edd905ce5e7 : FAILURE in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a49e3df605a40a2a0e4a4da6442b556 : SUCCESS in 9m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bab18323223847e2bb246be1de2b3fce : SUCCESS in 10m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7712a18067314b14b962eb4da6e4ac7f : SUCCESS in 14m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b210dc49f804699aa8186db0b05dc48 : SUCCESS in 7m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/8855ea1b264e416e8b505b2c7e3a7d50 : SUCCESS in 34m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/31dc261f8d9d4eb7bf97492eb9ec841f : SUCCESS in 52m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/874664bae688491fa2b2a554d20e0c04 : SUCCESS in 1h 24m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28f96f48881c46fe85082ee8ada74b54 : SUCCESS in 59m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dfb9e040e4f848c59bc4049cadbb073c : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/40743e2032384a9f99624356e342e531 : SUCCESS in 3m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/63d20d67bffd42358ed01631b638db59 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/05e1d774eb16448ebefe2a1b2755ebb0 : SUCCESS in 5m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/991f7b33a8f54e17a4ddcb307ea0e646 : SUCCESS in 8m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1b4ffa4db4604c528f8979a7748b1996 : FAILURE in 12m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eb816e164d66422fb1959ab0f4d35d45 : SUCCESS in 6m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9ccf2a40087f4340bfc109d050ad6065 : FAILURE in 1h 32m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d771d703eb6241578735428d03f797b7 : FAILURE in 2h 49m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c336e7cba6a84f2984e598210b662904 : FAILURE in 1h 55m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/25665aa6d87943d8948c373c899561fe : FAILURE in 22m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/55a8d1f121d74c2ea18347e555985e07 : SUCCESS in 18m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a855f4d8e0b34b96826228329848a940 : SUCCESS in 1h 01m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/65dc96dd660b42b18dbc21b39f257a0a : SUCCESS in 1h 26m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b766152dca32473ca1bf3d4ca5a6fcaa : SUCCESS in 42m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f00ee99dd018477399095438b464e90c : POST_FAILURE in 1h 56m 27s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a78e5ea6370047d3bfffadf929745ebb : FAILURE in 14m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8fb6312ba6424d1fa9b62cf3cb416628 : SUCCESS in 51m 43s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf6a3a95fe844e74900307a7174bbd3f : FAILURE in 1h 22m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/309863a188ab4c439cdcce9a47cd0045 : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":1},{"id":"81bc9ec5cfff5f26f8e2d9a10036888ace4815b9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-13 05:34:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1566/.","accounts_in_message":[],"_revision_number":1},{"id":"cc17d293ca590cf9fe7554bba0165295405e2d1b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-13 05:37:47.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1431/.","accounts_in_message":[],"_revision_number":1},{"id":"f202f9ecee9e68594b909a66f588ec2a73f456ac","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-13 05:58:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9df2ba47ce094ad69ac48d249d21bf84 : FAILURE in 42m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/02823d859796422d9b41590a09ac2122 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":1},{"id":"60fe4b8a928b80404f67dd2f33a82792678f70e9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-13 06:56:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/957190 : FAILURE in 1h 24m 28s","accounts_in_message":[],"_revision_number":1},{"id":"43e552d9f7ba55b2214899de89025a7c4bc62649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2025-08-15 20:34:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"64fe99b4f11cd25fddb22707cd0318dec79025b9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-15 20:34:15.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"cdb9003d5ce414622a133840345a1951292d7b78","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-15 20:43:19.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/185/","accounts_in_message":[],"_revision_number":2},{"id":"3a3b092688d742569d485f5f397e6ee883804712","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-15 20:55:52.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957190/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957190/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"132ef12fdaa45b4e3ee76e28f8342350ce8dba60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 21:08:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9939e48675c7459d8b06cae21ca5fec9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2b481d3a7e934ceba7cdfb5198a420ae : SUCCESS in 23m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c7226831fbd41ff98e703ea533168d8 : SUCCESS in 23m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1e062af3031c44bb8bfdbb4bbd089c0a : SUCCESS in 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb997ca2ca7449236d5330f87e9ce93b145f0b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 22:40:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11470ab83ff248309bb09d83059387f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e8fe73f113a4ed397d129530ff2480e : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6179977ec24d43a8822573a54b3cef3b : SUCCESS in 9m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ac1c89b370f41319af18d09d9523675 : SUCCESS in 12m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f51232bb5f944a5871b8032d752e3a3 : SUCCESS in 15m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f2f950584284451b42ad5b0ff3defce : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/07323fc49b5048ef995654557d746098 : SUCCESS in 46m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ae9a14f7872d4fe3a8fea95f38a02f64 : SUCCESS in 55m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/270df74059af4f8aa5296cdeefd3c8f5 : SUCCESS in 1h 36m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/065ec6cd42f5469999528d93ebb2cdea : SUCCESS in 36m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0c8774e222344d508747841bfe3ec05b : SUCCESS in 15m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/544d6d8ed59b4dba8df9ff39048f4436 : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/14796c32bc0a4dd69909b26d1c9778a1 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/682e53c9d9684117b5ccbd9fff1e77ae : SUCCESS in 6m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1ed4de16693746278375e84e64e8617b : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e79834f8ab3f4ae799e5664af3c93e87 : FAILURE in 23m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff3216009c8f4e4484a7ab661977da02 : SUCCESS in 3m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e97d9dc0bc3f4da3bacca37f7876b482 : SUCCESS in 1h 11m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a9da8331e36c4c1db32dbc10e96e0b1d : FAILURE in 1h 48m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1903fe26e68f47ea8c4801192e9fdcb6 : SUCCESS in 2h 03m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8d521232a9f444428d0c8fbe497a3fe1 : FAILURE in 13m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3c125dcf24a1453fbee5e3c3cbc0b6f1 : SUCCESS in 30m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c8f2ca8a87df4bba979c077fb9e9f706 : SUCCESS in 57m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7e1f207f3fa34fe8960809b9e03025a7 : FAILURE in 1h 50m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/94be73eaa0e0469aa729450fca8bf51a : SUCCESS in 22m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9c99962178c04287b0e7cf5212b70827 : SUCCESS in 52m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4d4954c5547e4813a6ad3e7f1125b0b0 : FAILURE in 11m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b4c71100f3f4cec9bc1dec0a7060a53 : SUCCESS in 56m 59s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45726c42488b4ac8965c1ef62733a965 : SUCCESS in 1h 27m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70d8fb45892042fca4e5b446d324c5a6 : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":2},{"id":"37fa3c3c45ab2994ed980c4993ed6c06b17f6465","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-15 22:59:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1584/.","accounts_in_message":[],"_revision_number":2},{"id":"05edb1d8a5a363eac6b6f1770cf7ad38f3d6a862","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-15 23:12:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/957190 : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":2},{"id":"33f5f07e1b9d5fae28bd12ed266b5530cc938b40","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-15 23:16:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1449/.","accounts_in_message":[],"_revision_number":2},{"id":"ab9867381768bd4464487a9d94b45dd23ed571d2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-16 00:34:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/957190 : FAILURE in 1h 25m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c59a18b2b19223e78d897a9355d63ad3c2756040","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-16 00:36:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8ba43c34e028440fa0b01c4abca14388 : SUCCESS in 1h 36m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e7902ee9522c4761b25cb02141193e82 : FAILURE in 1h 46m 33s","accounts_in_message":[],"_revision_number":2},{"id":"864433f9dc1f36bd457a50eebfbec2b82173835f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-08-22 16:49:39.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/90/957190/2/20250822153917/NEC-ISCSI : FAILURE in 10h 02m 28s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/90/957190/2/20250822153917/NEC-FC : FAILURE in 10h 10m 20s","accounts_in_message":[],"_revision_number":2},{"id":"810a963c351c4e39ef5c3956beb8d7a7bbf989a1","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-10-31 15:58:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74b0d46064351f14a2a3ea3392390c68b086aa0a","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-10-31 16:24:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7deaf15dbdc0b51696d426fae8a5a76d4f72a45c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-19 02:14:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a0e9edcf8ed8b42c237c2478353c01fe234d75e1","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-19 02:51:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver build/54185fd3be7c4dd887fa35aee12c8122 : FAILURE in 29m 25s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/957190","accounts_in_message":[],"_revision_number":3},{"id":"a7a4336f386e8e406f78cc13c9c0accf3e071275","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-19 02:54:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/b9aaba0fd3d54a26ac50d8ce475fc8f6 : FAILURE in 28m 59s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190","accounts_in_message":[],"_revision_number":3},{"id":"564b2382ce5f9e6ffd16b34816ee9d7a1c693dfe","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-19 03:01:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/e3f3ba10f4b94b349f69439af659d7cc : FAILURE in 38m 43s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/957190","accounts_in_message":[],"_revision_number":3},{"id":"0ed43c2a1861539efe9a65d1ed7249861b39f099","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-19 03:06:04.000000000","message":"Patch Set 3:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/957190 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"51c459133a54748b79f8eaa620bf96bce7a31050","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-19 03:16:54.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/02851dfc093c4d41b909c8d78a0649ab : FAILURE in 48m 05s\n- cinder-storpool-tempest-iscsi-multipath build/f5cfe0819c7e492d9cbc71e4f5a62409 : FAILURE in 51m 20s","accounts_in_message":[],"_revision_number":3},{"id":"ee21e19a77fda12a851a619eca01a7332b713778","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-19 03:20:14.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60 .","accounts_in_message":[],"_revision_number":3},{"id":"78f45ac3e4105605b13a7e1d2a4a09118676fcb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 04:46:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4f105a5f8884e2f90f67567f5f9e6b2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ce21e7a0f976470ea4d6acac9725682b : SUCCESS in 16m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dfb08a316603484287c83d4f327608af : SUCCESS in 16m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c527ea48d9de441b8d7f53e4bbb5c611 : SUCCESS in 23m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ea86bb311cf85417f42fa6042afab3f5b14a840b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-19 04:53:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4184/.","accounts_in_message":[],"_revision_number":3},{"id":"17a47dff546eb573d9c5473a16cbffd3cb6b498c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-19 05:02:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4008/.","accounts_in_message":[],"_revision_number":3},{"id":"6e38e377a83c0021c33b4cf17cab5fb948af6a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:13:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b56f2899068e4bb088e7798e75e5d949\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24967375745a48d89b38b5b906bb93a9 : SUCCESS in 7m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18739bb7844c4944865e1eef256e09f9 : SUCCESS in 7m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/473869ee8a604028b5cfdccfa684553e : FAILURE in 11m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/431eadd1ba1a4daca1ee4489bcb3b20c : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4e79ba4a34c431f89204698528a1e64 : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/98f13b3a6b1d441e917d1d9b550298be : SUCCESS in 1h 01m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/479c1ffcf43641d4926c25ef2c135a8a : SUCCESS in 30m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75874f76b9454b6bad92b3bf665f973d : SUCCESS in 1h 41m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23eba20d2cf7464a9bb6bdb8f1c86216 : SUCCESS in 33m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01fdc61d82d746dda893b14bc3d048f4 : SUCCESS in 7m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1f1e22f390304235a981094afdde83b0 : TIMED_OUT in 41m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/429cee714fd044e6a9310f094e9d68ff : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8fd8ba9ce4e1407ba10912926f1ec79a : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/73d0d27ec5a048ea9879f62d7f240a60 : SUCCESS in 6m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e598b301dea94c5cb38184ab3d064c86 : SUCCESS in 9m 07s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/bc0dce3d9f8f40459463fdbe28fe7cc9 : SUCCESS in 11m 11s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a52d69310a144da5b47ab6fb98d3c0da : SUCCESS in 42m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/07955702479a452e8421b4d642d773bd : SUCCESS in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a394457b847a42a7ae94b2766a08fde1 : SUCCESS in 1h 40m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/99a832e441aa4936bc58b1d040e411ae : TIMED_OUT in 3h 04m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/762bd2fb0cf04c1694ccf5802f5591d3 : SUCCESS in 1h 44m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/17434b513457446a84e969f823a7d6f9 : RETRY_LIMIT in 2m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9d0c2ee6cd584d6c825861bd40d724a3 : SUCCESS in 28m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8f4be1e27e9e4e25b44b18e950a5ac77 : SUCCESS in 1h 17m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6875b4b994342ebbd9f338d991df8dd : SUCCESS in 1h 52m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/99975d4d69e24cb7b634c2de5866138a : SUCCESS in 41m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a39caeb3e1554a78838ffc1380c2c182 : SUCCESS in 1h 31m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/089ef4b772174f5f9476084c62041739 : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":3},{"id":"0aad4a07070bff8afd28b29433bef91ae55429a7","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-19 18:51:42.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cd3b9adb12eb5f27837c15f24ef212ee8e07efb3","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-05-20 04:48:44.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dbc759486ab5e3d161ff31a1a54d983965f223d4","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-05-20 06:19:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d1f778565ea85ad2fd9d9855acf088edbe2dbcd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-22 01:14:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"0a5ec1e6fdcbcd6f66fa4ba3a0a65db617dfc44e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 01:54:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6281baa206d9428095e34eef8a9a7b7c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/908702815b0b4932970c8df425078acd : SUCCESS in 22m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/382253adbc4d411294ffe3903954d0fd : SUCCESS in 19m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/146c660c16194316a6e985396b91388c : SUCCESS in 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"75b0bfc58bba01e7913350415166515634dc8cd5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-22 02:00:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest build/b5846017ff6e4f24855cf504aafb6a06 : RETRY_LIMIT in 24m 17s\n- cinder-storpool-tempest-iscsi-multipath build/ec4ec29f8d4a4dba9fb308e30d1dca68 : RETRY_LIMIT in 24m 02s","accounts_in_message":[],"_revision_number":4},{"id":"cd1f08f0a5208afb0e1331d0fd41e4a652eef577","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-22 02:11:02.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60 .","accounts_in_message":[],"_revision_number":4},{"id":"05bd2f6e40e7004dd0264f7e7793db671d004576","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-22 02:12:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/3a311ea4a54348f3a3001dcabb0830da : FAILURE in 28m 32s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190","accounts_in_message":[],"_revision_number":4},{"id":"63d322c523c97940a55e4315b5d30ddea4053bc1","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-22 03:17:53.000000000","message":"Patch Set 4:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/957190 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"bd85ee5f657a501f0ed0c89b5d730b84e160c58c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-22 03:33:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4218/.","accounts_in_message":[],"_revision_number":4},{"id":"eddfb70ecb51560482bd227bc40e2be5170a61bd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-22 03:36:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4042/.","accounts_in_message":[],"_revision_number":4},{"id":"e0db07859137a74243b65beb3d91f990ccf0da1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 03:46:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/905996e8ae894cffbeb0e9b857c508ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/170de68853894ac9aa93568f4b8ce089 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f38f27e54b45411e9f0f53347d1a8421 : SUCCESS in 11m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cabedd4f69ad4e9da76ade59394d2a8e : SUCCESS in 10m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/48e972191be34a3a88f2fe69f7a0c132 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b127bcfe47a548b29e830c599e29f2e4 : SUCCESS in 6m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/e3ddec67f9c6422b8650d5e06c18288f : SUCCESS in 1h 04m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/97a9d99260ca43949b30cab58e9fbf1e : SUCCESS in 36m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d6c4c290b9341c4a641f9e62ad96bb3 : SUCCESS in 1h 38m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/337eeb4597b44a00960568889b71c56d : SUCCESS in 45m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/781cfc0422514c8fbce70e84b58bff88 : SUCCESS in 7m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/17bc33717fe5450c89df7f3c87b08aee : TIMED_OUT in 40m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d5528713aa23421b88c83224ab3ff877 : SUCCESS in 4m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b4ed5d5532c74b1fa51882a8f3ca7148 : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/141c4a770a7b4a29953e115c145a907d : SUCCESS in 6m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/87b36db7620245f38ced94c5f2e82565 : SUCCESS in 6m 02s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b5457520d80949319706dcb659938423 : SUCCESS in 7m 56s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/14f3f1a6ee51483d95b4331b2fdc2d55 : SUCCESS in 41m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff5dec5ae6bf4d22bfa12cdd7abbb9e1 : SUCCESS in 3m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f5262423ce064eb0a53ee8b38b728696 : SUCCESS in 1h 28m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a230549814774cb9ae6f8524ae395273 : FAILURE in 1h 30m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2ff4bf8c26d14ad6bb555b3aadd1cae5 : SUCCESS in 2h 06m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/521a476267ea4f2b9e8b0c00c131a58b : SUCCESS in 2h 26m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c4a50811ad3b44809bbee4d0387146a4 : SUCCESS in 26m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/526abc9e99084032873469cbc7b5baca : SUCCESS in 1h 15m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7ceacf2de9c044b8a7f3e753ff2ee956 : SUCCESS in 1h 41m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b5cbaabfaf474b5696559e9531ec6086 : SUCCESS in 40m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f962cfa8970443e180318e355715876e : SUCCESS in 1h 40m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b000b87066a4bcc98b1353be0e2dcef : SUCCESS in 49m 12s","accounts_in_message":[],"_revision_number":4},{"id":"1b86b64fba33bf5cbafcd1dfb13789c324aae60a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-22 03:56:48.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_05_2026_03_49_49\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"405382a7d5d2ee07b5711f6e0a7bebd06091e675","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-22 04:52:05.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver build/1cc5dd104c034991aaee22f7b8ba37a3 : FAILURE in 40m 23s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/957190","accounts_in_message":[],"_revision_number":4},{"id":"0f780b026342bc4ab878a246e462a50b5a4eaa50","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-22 14:37:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/346c834651ff4904a76c229d5ae863b2 : SUCCESS in 50m 20s\n- cinder-storpool-tempest-iscsi-multipath build/f34198c0c0654ff6b84a82d60e7d06c1 : SUCCESS in 54m 39s","accounts_in_message":[],"_revision_number":4},{"id":"f62a355e7415a1551918231834106ebfd2b5a469","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-05-22 15:53:40.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a008bb8f0967c434b2cdb8ac23a4d5aac98fb22b","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-29 02:59:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2e1f4c15a8a5478a0ed60fa9f9cd8ce2b603191f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-29 03:00:19.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f35378c690613ab7ed1f02612f6cd8f0e2046c8c","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-29 03:03:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bacd28f0eb95b1126ef8cc3bc59a6778fb0b67ab","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-29 03:20:02.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest build/17d0456fc1ba4c6eb33759f98d9e498a : RETRY_LIMIT in 18m 26s\n- cinder-storpool-tempest-iscsi-multipath build/c95a076ada10454fba2c2ddd98f5f446 : RETRY_LIMIT in 18m 03s","accounts_in_message":[],"_revision_number":5},{"id":"9c2b6b95c966e7a58696f4f3ea77db1f06e4cdab","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-29 03:22:39.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/4a19461287db422baf8e088d31d0e02a : FAILURE in 21m 51s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/957190","accounts_in_message":[],"_revision_number":5},{"id":"4856b83adb317d0e4d21e861af96e843ffcc56af","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-29 03:41:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a86f713e3ad64603b5ecce5f528a0652 : SUCCESS in 40m 14s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190","accounts_in_message":[],"_revision_number":5},{"id":"58d09b5c096341b8ebc34653ae4e4082f5024031","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-29 03:41:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-fc-driver build/1a7b59769ae94dfea2eba24e35f86e04 : SUCCESS in 40m 54s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/957190","accounts_in_message":[],"_revision_number":5},{"id":"7e0c798eae59fa7e82e1d8cdb0fe9a586ebef2bc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-29 03:57:54.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60 .","accounts_in_message":[],"_revision_number":5},{"id":"d5c5cf01439ffd3b61f727822db16b85df3991ea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 04:07:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4101/.","accounts_in_message":[],"_revision_number":5},{"id":"528f1419c914bd22851fe484c9be94265051221c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 04:08:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4277/.","accounts_in_message":[],"_revision_number":5},{"id":"653b49165d3487430e0de6a69cc760ebb614d58c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 05:30:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e49c55215849411f8f76d9ff9ce84911\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33411ac6f35b426b91c0644cdf7d6e80 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be325bea80704733b2d4e6c37ea0301b : FAILURE in 8m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be10a0b999224de8b4def38bcf12eb8d : FAILURE in 10m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd078dc1859c4495a6f666b01aa16606 : FAILURE in 12m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9abec04d96a44269270912cfef947a0 : SUCCESS in 15m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/ccb0b2bb7e2444ad9159dd5fe751fec4 : SUCCESS in 1h 11m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c30017615e14ea09ec1eff884db121d : SUCCESS in 1h 07m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/176ec64a01304667b5d3491f892020e3 : SUCCESS in 1h 16m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96e15371cb044452954e9d18940a69fd : SUCCESS in 33m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2e4e75078914b4885f4c8e84c36f38a : SUCCESS in 8m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/249f6f14c8b04261b610b1bc6d3e99a9 : FAILURE in 20m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2ec908c09d0a4b9d89f82e7494c5a345 : SUCCESS in 7m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f17b73bdd7bf4b18a47d4740fc7f0332 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/82e66531f587486ca255ba4398c2202b : SUCCESS in 5m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ac0bdd918fd44473b663919dea476fd1 : SUCCESS in 5m 11s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/327e394c437049f6bfba8ff98e77af38 : SUCCESS in 11m 43s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2584ae06b0e44caa9b0de6a494efc0c5 : SUCCESS in 19m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/16395c4772a84edcbe05a57a63ad4335 : SUCCESS in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/febdb4ae9c0d4cdbb0faef28fa0b44a8 : FAILURE in 1h 48m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5a7f3e3ccb0f4deb98ef4eb93bd1e642 : FAILURE in 2h 20m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/843e9cf1dc1b46778459c48682aa9641 : SUCCESS in 1h 46m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/90fe9eda48c74b2f9371c3b945b42da4 : SUCCESS in 2h 04m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bfc888786f664c498ddf81e7db5a2a6b : SUCCESS in 28m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/71c5c6ca7bcc4b32b309e90c92443818 : SUCCESS in 1h 17m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/22f5d8fe00834a1a8a292b4bafd3a9c0 : SUCCESS in 1h 43m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0f8f167fbb554e45a7dd8b81bfb9f4d7 : SUCCESS in 23m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eb1cf1c6b73d459e91a46ef65eba4c4a : FAILURE in 58m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c968766b49d04bed9d3799665a446bb4 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":5},{"id":"dabca92decd3e0fada5f38c063706f36e1c575cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 07:38:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c1c0aefd874dba998f44490d3f9062\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/10f79aee1b62417abede42730ba7f3d6 : FAILURE in 23m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b2011bff4bda4c219ce353d81f9d9df9 : FAILURE in 19m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/19d110f7dabb4402bc6564931473a345 : FAILURE in 21m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2a9a28af510a283fdc26573c657ba342b5564731","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-29 14:14:35.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"907cb527b7e2b40dc0669961ec998019df18f132","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-29 14:41:16.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest build/5f0362aa92b34688b520dcea24ef396e : RETRY_LIMIT in 19m 03s\n- cinder-storpool-tempest-iscsi-multipath build/c5ce3f6fdc334e589013bfbbe856ac05 : RETRY_LIMIT in 19m 57s","accounts_in_message":[],"_revision_number":6},{"id":"b1240b88895e8d761cb4e28d95cd18ace03331ca","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-29 15:11:47.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60 .","accounts_in_message":[],"_revision_number":6},{"id":"5e147a2e79c7a9dfbadbb85a9550cd856a34f011","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-29 15:17:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/ab50d82294bd42d7892e5c05c9758676 : FAILURE in 22m 03s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/957190","accounts_in_message":[],"_revision_number":6},{"id":"64d1679b9edabd2853d7f3d50e2334e91b188624","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-29 15:41:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- hpe-fc-driver build/ece369ae898e41bb8b8440b3d9dedefc : SUCCESS in 37m 37s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/957190","accounts_in_message":[],"_revision_number":6},{"id":"ca1da094be89d5870559cc2ecc84d5f292b659cc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-29 15:51:58.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/ac0bc4604d3d4d4689ebb9798769af72 : SUCCESS in 40m 30s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190","accounts_in_message":[],"_revision_number":6},{"id":"20e425409dd5e5d948f4d7a2d67b1d9848fae1d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 16:50:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97872008129346a0afa3c4a2fce6e536\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe02fa4b68341379021f323f9ab505d : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33d4cb5af10f443eaeee545626abd0b3 : SUCCESS in 11m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf069edcfb584e0aa692396695ba5b81 : SUCCESS in 10m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4408b57f3fe54aa58224bf29e1a7e1f4 : SUCCESS in 15m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/535b022c130b4c208c092c35472b2002 : SUCCESS in 11m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/cb25f6f5d0be42dd9a6477d7483ac5e9 : SUCCESS in 57m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/27c3c17d4a584ab68ee5e423b91dd876 : SUCCESS in 33m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99cce63e786c4efeb2fbc96566643a60 : SUCCESS in 1h 20m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdaa4b7fb1ca498ba407104eedf0301a : SUCCESS in 1h 03m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acf5d289d65349d099ed95c5091968a8 : SUCCESS in 9m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d1ac728fe87f430eb0d94728af289fe4 : FAILURE in 25m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ca1d5c2f9ecc4696ae50121b3b576064 : SUCCESS in 2m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/24235c42d88e4c329a6d0a642912cf25 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c2e5d314ad2340ffa968e79dfb6e1bd7 : SUCCESS in 5m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ce0535ba4e8148de8bd7ef22a914b7b1 : SUCCESS in 6m 18s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/54dfe19f4d7f460490cd363672d4265e : SUCCESS in 7m 44s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/501dc576feb842fbaa1162aacee5158e : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bf66995ad2f74a7e9f247bf063b8ae15 : SUCCESS in 3m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/72da012ebb844e2ba808596e43402acd : FAILURE in 1h 59m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/257097d588474e9d99f981366fd068a4 : FAILURE in 2h 25m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a4f222ae5eb14bcbb771050f3cb4715c : SUCCESS in 1h 57m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7ea3cd471f8045f1992f81d879757847 : SUCCESS in 1h 54m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d3abe15c81564c0db51ddd750b6fe82d : SUCCESS in 36m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3aa8e68716194b61b8b9f09f7ca15a0c : SUCCESS in 52m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f07dbf06723644e7b7e62738d57d2bc0 : SUCCESS in 1h 33m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ff956630c3c745539fc6af2df8b85e2e : SUCCESS in 48m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3d8ba4bad5714f53afd7711c25b906bd : SUCCESS in 54m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df4e0939e2e444479598c19fbb945004 : SUCCESS in 1h 06m 11s","accounts_in_message":[],"_revision_number":6},{"id":"ae4b522beac4e1aa2b2780f144169783ac8cde20","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 16:59:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4106/.","accounts_in_message":[],"_revision_number":6},{"id":"083c4087f5b718f1af6b0d92c609e35405cc6265","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 17:03:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4282/.","accounts_in_message":[],"_revision_number":6},{"id":"870ca187e9295c934bb980f0456a76767be78c74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 19:52:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e78ca2a7b00f40c784030d8cfd3d9edd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/571598971a8b49b689da152e244a95d1 : SUCCESS in 17m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c56a576bd05144e78f95fde876624489 : SUCCESS in 14m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/481c93c1f4234ecf8c187ff640924573 : SUCCESS in 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0e2b2ca792f245be4a01ecacbea087ecbde2624d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-06-03 01:20:54.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"2e21c126b3622695dbee17cd0638de4aa2aa98e1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-03 02:01:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/890999626bef4a6daea0cfdd64e8fb6a : SUCCESS in 39m 57s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190","accounts_in_message":[],"_revision_number":7},{"id":"596320be6040d3b4e5ff03e292df44951315e689","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-03 02:02:26.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hpe-fc-driver build/34ab090fdb5e43f686f25806251cbb35 : SUCCESS in 41m 09s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/957190","accounts_in_message":[],"_revision_number":7},{"id":"1f3e403beea5219e37829d4b109240db4b72ca6d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-03 02:08:42.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/c3785266a5e34db1bfa6d7cf2383add3 : SUCCESS in 47m 28s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/957190","accounts_in_message":[],"_revision_number":7},{"id":"0abc0a6872be3cbf5d04d7bba07bd90c79f7038e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-03 02:17:32.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60 .","accounts_in_message":[],"_revision_number":7},{"id":"d132c8f20da5f861517bd9ea8b3cf5e12d6e5a40","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-03 02:39:15.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest build/27dfd4c9bf59449c8d0b8f7205991403 : SUCCESS in 52m 36s\n- cinder-storpool-tempest-iscsi-multipath build/4380b9fbfab445acb90c586efcb0172c : FAILURE in 1h 00m 44s","accounts_in_message":[],"_revision_number":7},{"id":"2379562255955b0d23b5e49274d2c97b948611d4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-03 03:00:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4155/.","accounts_in_message":[],"_revision_number":7},{"id":"e07030f2d5dc85ddaf44681b066977638a2783a3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-03 03:15:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4331/.","accounts_in_message":[],"_revision_number":7},{"id":"cf58518ab2ffa8114ee99479ac4934fb3533e5e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 05:09:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2af13e2ccf454dbe8dec35149482bccc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9be5ffc24b84d7caddb88600f94f8d6 : SUCCESS in 6m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d49b58bf9ca475cab14fa613bef966f : SUCCESS in 10m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0308f98d0dee4b6783258680f3082cef : SUCCESS in 9m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cdc31c708dc84f0cae58308cb7d87fa1 : SUCCESS in 15m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0518253c885b4f74bfdfa31956bba7f4 : SUCCESS in 16m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/a3a05a11dbac4676b627633392410eb0 : SUCCESS in 1h 01m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/19e80a48a25146dc8531f0c326dea79c : SUCCESS in 1h 08m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce8a5046eaf2401c84b1f8228663788b : SUCCESS in 1h 41m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abf29875e75f4ec19442fe67841b6e91 : SUCCESS in 1h 05m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45bbf840eda0434a9b02f0f43e7ba60d : SUCCESS in 5m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b2d03459ca024cd7bcbe209063c6facf : SUCCESS in 16m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d5323207fcfa487dad7768afd4d49a8b : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6ea84e88b5844165844cf838d1cdd898 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a11467720f404b86bc5c69300fa5b56e : SUCCESS in 6m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e060b9037d5d42f69e07c2fb59b62b8c : SUCCESS in 6m 57s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5b4c0397896c4777b83f4a937c5a109a : SUCCESS in 7m 47s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9de22af344d248dda5ebf71f9fc325fe : SUCCESS in 27m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/303cefcd09754635837aee61fbf1d269 : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/14f7427a8945444894253e2a6aa06261 : FAILURE in 1h 47m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/69957440c7f44133803ccca207edf006 : TIMED_OUT Host unreachable in 3h 43m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f1e2ea53d5a04f2296d4c40497e4b2f3 : SUCCESS in 1h 58m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/daa701fd2f524c67b229d4779526745b : SUCCESS in 2h 28m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/70ed8c6e320f47ec8fcc1bfd54990540 : SUCCESS in 31m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d8f32e9cf6b6421d8c9457946adfe375 : SUCCESS in 1h 17m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5ad04fefd3f54df787dc391407ceaca7 : SUCCESS in 1h 55m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/78c2fc5f36ef4fb09b0905dad7f41cc8 : SUCCESS in 23m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2520178b19ae47c485e1864ec642648b : SUCCESS in 1h 13m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b568101dc30462b8367e5c87ff1720a : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":7},{"id":"cdd89d4be91d58b9a37fbae81aadc5b370d17071","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 06:15:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a90746301bd242f29f7ed57f9cb07256\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0669fc4aed224093831c4414044b1fd6 : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d84156df7d4a4461bb9846b113d9fbf2 : SUCCESS in 13m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0c06e5102e104096a0ede141e146a555 : SUCCESS in 29m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cacf4b466488f24136a0ca4e5fe0b1753b1cc034","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-06-12 02:11:59.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"1baa0774aeaef9c3eb1cee7648b9188ac8be38ac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-12 02:34:53.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9481db08ca0a4f7b8fe4c149c3d0d708\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b31fcd51f74f43d0a6ea7aa313609846 : FAILURE in 21m 19s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fbfa26e21b4b4ff9bb1e4ae3ad23c32c : FAILURE in 21m 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d6cddb4fad994505b0ec96cdc322f5b0 : FAILURE in 22m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ab2cf4d586a248e89d46d9e3168b99d8 : FAILURE in 21m 39s","accounts_in_message":[],"_revision_number":8},{"id":"41bd0aeafd2da3c79baa9632d6cdaeb6de42d279","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-12 03:12:50.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60 .","accounts_in_message":[],"_revision_number":8},{"id":"fdf95f40c44aca07895d565213527e76a66afb54","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-12 03:30:37.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a4a15d1a43a04443be244dcec58557ca : SUCCESS in 40m 20s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190","accounts_in_message":[],"_revision_number":8},{"id":"81887c6e15747880f8d6d7a4ff829c6d7b357cb2","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-12 03:42:57.000000000","message":"Patch Set 8:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/153 : SUCCESS in 1 hr 30 min\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/957190","accounts_in_message":[],"_revision_number":8},{"id":"865570b7f1949d3613192b7623389a79195edd05","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-12 04:07:08.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ead1e85cedd749bca9a8326a36020304 : SUCCESS in 48m 27s\n- cinder-storpool-tempest-iscsi-multipath build/559bb822671c46c39c369c49a618f3cd : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":8},{"id":"e9747e4cbef65cbbf0b3dff8a20a92835f4d6094","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-12 18:52:44.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4465/.","accounts_in_message":[],"_revision_number":8},{"id":"39bca3afe87aff181daf6e07f4df8db2a0ec1204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-06-17 01:29:48.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3592ea7b5c0f066a6d0eb2178a269704d10ad8ac","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-17 02:08:42.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hpe-fc-driver build/396b9f6397fe48e897f90224fb484f50 : SUCCESS in 38m 20s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/957190","accounts_in_message":[],"_revision_number":9},{"id":"647652e83dbc19926a5e14e6e7b241481fca944e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-17 02:10:00.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/7146930ec3f545488f8a1461d4b68e9f : SUCCESS in 39m 45s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/957190","accounts_in_message":[],"_revision_number":9},{"id":"9b8231e91601c8a9c59f773bcd3f5df019db913d","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-17 02:15:38.000000000","message":"Patch Set 9:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/472 : SUCCESS in 45 min (This includes wait time)\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/957190","accounts_in_message":[],"_revision_number":9},{"id":"7f5f6fa6306517794be7089dc5a21edd860beda6","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-17 02:16:04.000000000","message":"Patch Set 9:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/123 : SUCCESS in 46 min (This includes wait time)\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/957190","accounts_in_message":[],"_revision_number":9},{"id":"dbf64b5292d6aac6d8c3db391982ee0c6a2f0328","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-17 02:16:09.000000000","message":"Patch Set 9:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/209 : SUCCESS in 46 min (This includes wait time)\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/957190","accounts_in_message":[],"_revision_number":9},{"id":"ecce7d9bf7f563d00c275e638d495ef2aaa3871d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-17 02:17:13.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/b782c44a264347b08ddd9555041a0ebb : SUCCESS in 46m 54s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/957190","accounts_in_message":[],"_revision_number":9},{"id":"6a2a76bc1cf1681120fb0d6536feacfee648a3da","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-17 02:27:07.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60 .","accounts_in_message":[],"_revision_number":9},{"id":"c026cc5e75234e044abb1daf75d7b9a710fc697f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-17 02:32:45.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest build/1df5ff24cdbb4b178491dbf5aeb2fd6a : FAILURE in 58m 55s\n- cinder-storpool-tempest-iscsi-multipath build/b11ed816c247451ebd39067664d8db7a : FAILURE in 1h 01m 39s","accounts_in_message":[],"_revision_number":9},{"id":"9c78b8fa3944e22237b1b8c023f5d2e80fc7d1af","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-17 03:13:50.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/14ee7ae3052349308df82aa67cfbe351\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d127dd58afe04259a2ec67e6e3d7b081 : FAILURE in 36m 29s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a7f08d1acfa84f438867c581810fa0b7 : FAILURE in 36m 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3c687b25db3e4582bb2880f174d0eb00 : FAILURE in 1h 43m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/831bcf206fe24f06af77f34fd07d54b0 : FAILURE in 35m 54s","accounts_in_message":[],"_revision_number":9},{"id":"ed2761fe7f6a860587a06e00c721ac8044ea1409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 04:01:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18fa73ae68e34c5bbf0ef023add445c2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f965c7a018774483b41404fe5062cf80 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3960143e9f5b429892ac69eee79be8d5 : SUCCESS in 10m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26efd240b2b440adb3ba0bd3991db51b : SUCCESS in 10m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4ef5e68d2cf3440bb56a06f3eadd30d0 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba7caa409e14a8ab213b2aa1485540f : SUCCESS in 15m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/05cccb886d554109a25bea02a89534a9 : SUCCESS in 1h 08m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b5c495034bb544e68c22a75263fef10a : SUCCESS in 1h 04m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c7163ac328b24d9ca58a0fbfef17279b : SUCCESS in 1h 50m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f27f63c5bb0b446ba896575f8c40f694 : SUCCESS in 1h 09m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48b109b93b04409b81c0b64a4d17a537 : SUCCESS in 5m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d759a72bad9e486bb4f4ff7d408bce2f : SUCCESS in 20m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b187e2f4b5f340ff9711f0fbabe14802 : SUCCESS in 3m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a782b4b793a477884058eec07f44d25 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f2644f9169964ba18eaa7bab9ad65f85 : SUCCESS in 5m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5be1f1b9d1ce49c099220708ee768ae3 : SUCCESS in 5m 34s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c830f1197efc4b65b5680c7bfebc30bc : SUCCESS in 8m 05s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eb23d4aca274411497f4b350c5bdc71a : SUCCESS in 46m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/51c03b51aad2411c982caef255f9f3ae : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/02591361d954467da628b37ad5009147 : FAILURE in 1h 50m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/58b9dd3a4cce4974af755ae18c6a6a3a : FAILURE in 2h 02m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bc6149bb81e7491296974690ca8ee0cf : SUCCESS in 2h 04m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8bb5d89cfe294da98ff26e6d799b868b : SUCCESS in 2h 01m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e972b880cf5640cdb23e41d7992ce543 : SUCCESS in 37m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0a765ec710ca4504a0ce52bfe56922ff : SUCCESS in 1h 10m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/08a4e72bd31744e89c85a6c00753419b : SUCCESS in 1h 41m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c09960196cab4e479590409036ea66cc : SUCCESS in 45m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ee939d1ead034a2db9bfedd60ebcb9af : SUCCESS in 2h 30m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79e549a657c443f59210e363d80d192c : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":9},{"id":"9eba3c154a4f7632f826f2de886d466e320a4435","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 18:30:21.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2be9a39bdf7d4910a438dc6fb1238996\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e493b371fba647a6af20ba7c14a7ab41 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8811242d28fa43a19875cbc3b3c6327b : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8d2f96612efa43f4994f50f6a848878e : SUCCESS in 16m 22s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"93cc89477f437a381f34157ab9723cb4c7b5e7d6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:49:39.000000000","message":"Patch Set 5:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/2002/ : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"adebbf9de0e0368af482b33432ad85062eed9bf2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:59:43.000000000","message":"Patch Set 3:\n\nBuild Failed","accounts_in_message":[],"_revision_number":3},{"id":"b26ce930b8706d20885d08a7dd28cb14c0d1b209","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:05:07.000000000","message":"Patch Set 4:\n\nBuild Failed","accounts_in_message":[],"_revision_number":4},{"id":"4d3c09599a3e920d6ec6673668a14ec31f312d00","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-06-24 18:18:19.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/90/957190/9/20260624212028/NEC-ISCSI : SUCCESS in 5h 32m 12s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/90/957190/9/20260624212028/NEC-FC : SUCCESS in 5h 57m 49s","accounts_in_message":[],"_revision_number":9},{"id":"c67092e45e63a46b4008fb9d64bb001bbf7bf795","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-12 10:39:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4289/.","accounts_in_message":[],"_revision_number":8}],"current_revision_number":9,"current_revision":"65f82e13f1eb3beb1b9fe2015b4b62ebcd1ab3bc","revisions":{"7a32b9d92bb9e5aba8e40ab8dae3bcad611278d5":{"kind":"REWORK","_number":1,"created":"2025-08-13 02:11:47.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/90/957190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/957190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/957190/1"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2025-08-13 02:09:53.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2025-08-13 02:11:26.000000000","tz":-300},"subject":"Add additional rbd options to improve download speed.","message":"Add additional rbd options to improve download speed.\n\nCloses-Bug: #2120509\nChange-Id: Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7a32b9d92bb9e5aba8e40ab8dae3bcad611278d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7a32b9d92bb9e5aba8e40ab8dae3bcad611278d5"}]},"branch":"refs/heads/master"},"d75a557c6525922a5514d750f0307d9bc71e4cd8":{"kind":"REWORK","_number":2,"created":"2025-08-15 20:34:06.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/90/957190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/957190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/957190/2"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2025-08-13 02:09:53.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2025-08-15 20:33:54.000000000","tz":-300},"subject":"Add additional rbd options to improve download speed.","message":"Add additional rbd options to improve download speed.\n\nCloses-Bug: #2120509\nChange-Id: Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d75a557c6525922a5514d750f0307d9bc71e4cd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d75a557c6525922a5514d750f0307d9bc71e4cd8"}]},"branch":"refs/heads/master"},"6573cd71c7902d575ea84c43aec434d3deef6b18":{"kind":"REWORK","_number":3,"created":"2026-05-19 02:14:33.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/90/957190/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/957190/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/957190/3"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2025-08-13 02:09:53.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-19 02:14:13.000000000","tz":-300},"subject":"Add additional rbd options to improve download speed.","message":"Add additional rbd options to improve download speed.\n\nCloses-Bug: #2120509\nChange-Id: Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6573cd71c7902d575ea84c43aec434d3deef6b18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6573cd71c7902d575ea84c43aec434d3deef6b18"}]},"branch":"refs/heads/master"},"10f03c3a57e06fb0c2b490a2671c195abe29db09":{"kind":"REWORK","_number":4,"created":"2026-05-22 01:14:32.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/90/957190/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/957190/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/957190/4"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2025-08-13 02:09:53.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-22 01:14:21.000000000","tz":-300},"subject":"Add additional rbd options to improve download speed.","message":"Add additional rbd options to improve download speed.\n\nCloses-Bug: #2120509\nChange-Id: Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/10f03c3a57e06fb0c2b490a2671c195abe29db09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/10f03c3a57e06fb0c2b490a2671c195abe29db09"}]},"branch":"refs/heads/master"},"6e0fc02a28b3c484447eabcb20989092ad59f37f":{"kind":"REWORK","_number":5,"created":"2026-05-29 03:00:19.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/90/957190/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/957190/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/957190/5"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2025-08-13 02:09:53.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-29 02:54:14.000000000","tz":-300},"subject":"Add additional rbd options to improve download speed. For some rudimentary benchmarks, see: https://bugs.launchpad.net/cinder/+bug/2120509","message":"Add additional rbd options to improve download speed.\nFor some rudimentary benchmarks, see:\nhttps://bugs.launchpad.net/cinder/+bug/2120509\n\nCloses-Bug: #2120509\nChange-Id: Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e0fc02a28b3c484447eabcb20989092ad59f37f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e0fc02a28b3c484447eabcb20989092ad59f37f"}]},"branch":"refs/heads/master"},"b64f4b557d9775d78299a12b05bf4ef70424a04a":{"kind":"REWORK","_number":6,"created":"2026-05-29 14:14:35.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/90/957190/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/957190/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/957190/6"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2025-08-13 02:09:53.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-29 14:14:26.000000000","tz":-300},"subject":"Add additional rbd options to improve download speed. For some rudimentary benchmarks, see: https://bugs.launchpad.net/cinder/+bug/2120509","message":"Add additional rbd options to improve download speed.\nFor some rudimentary benchmarks, see:\nhttps://bugs.launchpad.net/cinder/+bug/2120509\n\nCloses-Bug: #2120509\nChange-Id: Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b64f4b557d9775d78299a12b05bf4ef70424a04a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b64f4b557d9775d78299a12b05bf4ef70424a04a"}]},"branch":"refs/heads/master"},"eface3b69c05670864c4bbacd95cc118f9152f89":{"kind":"NO_CHANGE","_number":7,"created":"2026-06-03 01:20:54.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/90/957190/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/957190/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/957190/7"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2025-08-13 02:09:53.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-06-03 01:20:20.000000000","tz":-300},"subject":"Add additional rbd options to improve download speed. For some rudimentary benchmarks, see: https://bugs.launchpad.net/cinder/+bug/2120509","message":"Add additional rbd options to improve download speed.\nFor some rudimentary benchmarks, see:\nhttps://bugs.launchpad.net/cinder/+bug/2120509\n\nCloses-Bug: #2120509\nChange-Id: Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eface3b69c05670864c4bbacd95cc118f9152f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eface3b69c05670864c4bbacd95cc118f9152f89"}]},"branch":"refs/heads/master"},"83f00a728d57fb6b2ae494be0fabc2461de78145":{"kind":"REWORK","_number":8,"created":"2026-06-12 02:11:59.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/90/957190/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/957190/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/957190/8"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2025-08-13 02:09:53.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-06-12 02:11:39.000000000","tz":-300},"subject":"Add additional rbd options to improve download speed. For some rudimentary benchmarks, see: https://bugs.launchpad.net/cinder/+bug/2120509","message":"Add additional rbd options to improve download speed.\nFor some rudimentary benchmarks, see:\nhttps://bugs.launchpad.net/cinder/+bug/2120509\n\nCloses-Bug: #2120509\nChange-Id: Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83f00a728d57fb6b2ae494be0fabc2461de78145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83f00a728d57fb6b2ae494be0fabc2461de78145"}]},"branch":"refs/heads/master"},"65f82e13f1eb3beb1b9fe2015b4b62ebcd1ab3bc":{"kind":"REWORK","_number":9,"created":"2026-06-17 01:29:48.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/90/957190/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/957190/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/957190/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/957190/9"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2025-08-13 02:09:53.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-06-17 01:29:40.000000000","tz":-300},"subject":"Add additional rbd options to improve download speed. For some rudimentary benchmarks, see: https://bugs.launchpad.net/cinder/+bug/2120509","message":"Add additional rbd options to improve download speed.\nFor some rudimentary benchmarks, see:\nhttps://bugs.launchpad.net/cinder/+bug/2120509\n\nCloses-Bug: #2120509\nChange-Id: Ia276b47a9be213b44f8b4614a7ba2e30e2c78c60\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65f82e13f1eb3beb1b9fe2015b4b62ebcd1ab3bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65f82e13f1eb3beb1b9fe2015b4b62ebcd1ab3bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
