)]}'
{"id":"openstack%2Fcinder~957450","triplet_id":"openstack%2Fcinder~master~I385940e8511750d7b0659ced31ea268f0bc8ce29","project":"openstack/cinder","branch":"master","topic":"fix-netapp-cinder-package","attention_set":{},"removed_from_attention_set":{"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2025-10-31 20:54:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I385940e8511750d7b0659ced31ea268f0bc8ce29","subject":"Fix the name of a core cinder package (netapp)","status":"MERGED","created":"2025-08-14 16:31:42.000000000","updated":"2025-11-27 12:49:55.000000000","submitted":"2025-10-31 20:54:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957450-fix-netapp-cinder-package","meta_rev_id":"d71f0f09ce9fdf5015483bc6f8e2344a287900f7","_number":957450,"virtual_id_number":957450,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-31 20:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-08-26 19:40:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-08-26 18:54:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-31 16:09:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-31 16:09:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 16:31:52.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-14 16:31:53.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-14 17:39:49.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-14 17:47:28.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-14 18:42:06.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-14 18:45:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-14 19:01:23.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-15 01:27:43.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-08-15 05:10:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-21 20:26:07.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-26 18:54:40.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-08-26 19:40:59.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-10-31 16:09:25.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2025-11-27 12:49:55.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"}],"messages":[{"id":"72dfb97410d19ec9bb62abdfd82053cca2ac4474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-08-14 16:31:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0a7cc2a0cdfe0894fc1207853cd58a9d5e58708","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-14 16:31:52.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/178/","accounts_in_message":[],"_revision_number":1},{"id":"9725cdc4ad9c33954c5e1d0ae19d710636acf33a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-14 16:31:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d817a9e470d284659df7c0b3713e09cc9c1f048b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-14 16:45:37.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957450/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957450/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"0874f693746551820d25682bdcc7765a98feb683","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-14 17:39:49.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1577/.","accounts_in_message":[],"_revision_number":1},{"id":"1af026182f60ad269ebc6fe89769ce3b803e9ee0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-14 17:47:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/957450 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c7519c1cb09afc5b39d94d8f8fd266b35685ff27","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-14 17:52:41.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1442/.","accounts_in_message":[],"_revision_number":1},{"id":"3ec8a3dfe0dec978fbb8785011e21e0abbc5dc61","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-14 18:42:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/957450 : FAILURE in 1h 24m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a8a6b200f73476bc1eb7fa0e857c1135aee5dbb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 18:45:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beb467b922734558b60ecdba93c57b9d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dbd209b71e42459f9af9d49f6e7bf5e2 : FAILURE in 29m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/edf3f0907dfd4b02848009dd6e4084a6 : FAILURE in 29m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7b16bd644ffd4567862a184e51a66f98 : SUCCESS in 35m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b848531ec3ab8e732cfab734b5cf29b884e2d9fb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-14 19:01:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fab7aae0cd7943f9878c0ab86250f521 : SUCCESS in 50m 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9ec8f0eb9ae0472a8a5f7e49bafbb0ae : FAILURE in 55m 46s","accounts_in_message":[],"_revision_number":1},{"id":"42809bc5be14cad8a7ed861fec80d4af255e9b05","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-15 01:27:43.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-50-957450-1/FC/ : SUCCESS in 2h 43m 10s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-50-957450-1/iSCSI/ : SUCCESS in 8h 52m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"703896a22c6f1ac33675489a5c89c1a3712abe55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 05:10:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c45846c5bc640e4b5f149b0a0441d90\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/790885f77e3743d395d4509c0d9464d6 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1de2fe953dc64469a7b081de60a2b03d : SUCCESS in 8m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5277706d2ee4a0cad1126d1b9f28050 : SUCCESS in 7m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/208a926fd4f3402fb9b3745634c2d5f5 : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/774db1287a9b4a20bac149cb50ad6f74 : SUCCESS in 6m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/c1ef801517a546d1be2b287c72762bed : SUCCESS in 37m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd098b58b9164f98b20ae58c9955547f : SUCCESS in 32m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/720a8c97beb848ae9a5b37d2b9b041d2 : SUCCESS in 1h 25m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1068201760054b77a010c0cfab2617fe : SUCCESS in 32m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ee68d8da3d5646439499215e722b5213 : SUCCESS in 16m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e4f7b0aa620541af8c1759ccbfe10805 : SUCCESS in 2m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ec491a3e64da4ab8872fcb6b1a407642 : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12406842a74e48d6a5398d80f9b8bbac : SUCCESS in 3m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1e52e587d77e4a1ba9ae2bf8261a9a63 : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1cc1db19477a4dc58eb0ea408f8f3cf4 : FAILURE in 10m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89594992774e46069d2084de7beb1b5b : SUCCESS in 3m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/be3f910cfbfb475d99e929c242301432 : FAILURE in 1h 15m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/51caf4186a3f496e8fd9a5bd6f03f835 : FAILURE in 2h 17m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e5fe6445d888408196b74557eb735bfb : SUCCESS in 2h 16m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/83b6c354a1534b22b01aa7de473ef544 : FAILURE in 10m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/29cab1d43cfb4aa1ae8f583bcda60834 : SUCCESS in 17m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c56c5a39c7c2429c9a5c19351d70cff5 : SUCCESS in 40m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/60a2041620014c909ce32605c0340da7 : SUCCESS in 1h 32m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ea492a06a23848a0abb79b12fd2f75ea : SUCCESS in 31m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6ea0ab6659b54b6cbff3d04267f17d17 : SUCCESS in 51m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/106739d3b8f14fcbb3eb4db2a9c8fbae : FAILURE in 10m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/47a406fb780e40179eb0160bc95fb9a8 : SUCCESS in 54m 38s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78f5a61365c34aebb7b567947d6eb61f : SUCCESS in 1h 35m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fecc3dc0d8b341b5aeae82f167a29a06 : SUCCESS in 32m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2f91fb5c0792b3635070f5fc27b26887a5bf6fcc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-08-21 20:26:07.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/50/957450/1/20250821190721/NEC-ISCSI : FAILURE in 9h 58m 10s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/50/957450/1/20250821190721/NEC-FC : FAILURE in 10h 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5181a041774f6e25949b0c5ed1e5074446a399fe","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-26 18:54:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae66fedf5b171738051f291b6e28460479c79af9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-26 19:40:59.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8d9809ed31e301379a397b8f30bbd9caab7d875","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 19:54:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a865a93255244fe9a6f52a39a2a43601\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91e3bd60c45842e8af246007da2066ff : SUCCESS in 22m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4618dfb4fb7d4911b05e23e18b7ed949 : SUCCESS in 24m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/24887790c0b54686a305a797b26e1493 : SUCCESS in 30m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28d86f9d9007d6015316d2db710bbf620ffd9545","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-26 21:16:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2e4b03778220400882893672f3bbe03d : FAILURE in 45m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f369c333715e4c629c6a2d31ee88cdc3 : FAILURE in 1h 06m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c1527a2419bb0d8e481cd5948e110ed75d719a22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 21:39:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c9aecacb9b8419f9b90dc1db0bd709b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ec0f9ffb8f7491c8c37764636f31f9d : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f493864c92d24deebcd8f09cd61b4809 : SUCCESS in 12m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/221fda59fad54f0aa3cbdb11c5d0526a : SUCCESS in 11m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cbf95d66f53949088baf7b117d27660f : SUCCESS in 14m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c441a6d38244a54870a3363bfa8ed6f : SUCCESS in 7m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/5d3a4609e16b4a18b1e98da94fdc33f7 : SUCCESS in 57m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2b892858f72f4ed0b0b8ef049823fe7d : SUCCESS in 34m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/36efd0677eb742799cffaae295994364 : SUCCESS in 1h 19m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cd9f836039242658c91255ded8aa5e2 : SUCCESS in 50m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/591a312475e74c6abf4f6a7854f8cb6a : SUCCESS in 16m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d6c2bea10fcc47d2ba98742606cfdd7d : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf866ac06f4f4e90a93f853fab3949ae : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e19ae980232745fe917ffe108b44eb6c : SUCCESS in 6m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/497ea9847efd497e913de0491aa22017 : SUCCESS in 10m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/36ed2b85d3e344f9bc8f8edd1da3a604 : FAILURE in 37m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7f6fd6d94db8448e864103bc1cb36d4d : SUCCESS in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a09edc080b0b42afb22b1039ef4769da : SUCCESS in 1h 47m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/022615eaef384e919e72a29064fe5891 : FAILURE in 2h 39m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2262b5bcd51e4434a02679f273afa6b1 : SUCCESS in 1h 54m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/db6a5cdaf8d44cfea131df91b4d16ea9 : FAILURE in 12m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6eb511d99e414d38a8b21a4af31c8df7 : SUCCESS in 35m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf096128dbca4f2697f129d774b100dc : SUCCESS in 1h 08m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/588b55c467bf4e5d9d0ad5afdeb87585 : SUCCESS in 1h 34m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/682ca9f296324053bf364e724daf6cac : SUCCESS in 23m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ffe0e9635a6943b1a605262fa8439dec : SUCCESS in 1h 33m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/15fde800d5c948f1ba806f46567a754a : FAILURE in 20m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2931d879fc9947adb252691b4a70cc46 : SUCCESS in 1h 17m 51s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4dc3090b88b744d48abef163d1ff2242 : SUCCESS in 1h 32m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7072f2fc2094b418d3e3371baab30b4 : SUCCESS in 1h 01m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6c38d9ee2da05cfaec3f16fd9989f298a9a57d67","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-10-31 16:09:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4239d7143c9985fdfcc2e96dcd0fc3ff7797ff7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:10:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1f17dfc588bd3b88d934cef50049ba923cdfda3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 20:54:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52c94abb297e40c5b20ace04eb4e8fc2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1371b1d2501944f795dc0516f1c3dd1c : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba8d2df3496a42feb7ef08a14c7cff7b : SUCCESS in 7m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9f5085d80c84df5b9548ac6dd888af2 : SUCCESS in 14m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c22acca19d475bbf74f209bfa13c17 : SUCCESS in 18m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/4a5b29d7a5e24dbd90dba862f195aa57 : SUCCESS in 55m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6c9855ec54ba40adb4617e0d518f0eea : SUCCESS in 34m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17c5ffb04b0f49a4b8fb013a0090e4b6 : SUCCESS in 1h 27m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10b87fbd2f564917a9dae039b8994a94 : SUCCESS in 55m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c4bf25a2b6a54c3d93713f75de7a6746 : SUCCESS in 42m 05s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ed210f8fb2b410fa79fea90cece49ce : SUCCESS in 1h 59m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/796cdcad81694c8aa922eac4f97406bc : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fd879910a413fb14cc81bcf46514e71cf2876ea6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 20:54:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e374d6b0ebed34719772aef3a4cc66ca8feb2997","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 20:56:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a1913bfda174d878eb1327a56207476\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c6383bbfc3458285b745008213f7e6 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d71f0f09ce9fdf5015483bc6f8e2344a287900f7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-11-27 12:49:55.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/50/957450/1/20251127173135/NEC-V-ISCSI : FAILURE in 3h 55m 33s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/50/957450/1/20251127173135/NEC-V-FC : FAILURE in 4h 18m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2bbfad0c412992d8f302c456f6e5bc83d31d810","revisions":{"a2bbfad0c412992d8f302c456f6e5bc83d31d810":{"kind":"REWORK","_number":1,"created":"2025-08-14 16:31:42.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/50/957450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/50/957450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/957450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/957450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/957450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/50/957450/1"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2025-08-14 16:31:19.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2025-08-14 16:31:20.000000000","tz":120},"subject":"Fix the name of a core cinder package (netapp)","message":"Fix the name of a core cinder package (netapp)\n\nThe package has been python3-cinder since the switch to python3.\nChecked on Ubuntu (since focal), Debian (at least since bookworm),\n*EL (since 8), openSUSE (the OBS packages at least from train,\nif not earlier),\n\nCloses-Bug: #2120672\nChange-Id: I385940e8511750d7b0659ced31ea268f0bc8ce29\nSigned-off-by: Luigi Toscano \u003cltoscano@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2bbfad0c412992d8f302c456f6e5bc83d31d810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2bbfad0c412992d8f302c456f6e5bc83d31d810"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
