)]}'
{"id":"openstack%2Fcinder~957891","triplet_id":"openstack%2Fcinder~master~Ibfbcff01c761a569587e453c952ae10bbee30fbb","project":"openstack/cinder","branch":"master","topic":"new-location-api-support","attention_set":{},"removed_from_attention_set":{"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-09-04 23:23:54.000000000","reason":"removed on reply"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-08-19 17:02:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"25402":{"account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"last_update":"2025-08-20 06:18:24.000000000","reason":"\u003cGERRIT_ACCOUNT_25402\u003e replied on the change","reason_account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-09-15 20:09:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibfbcff01c761a569587e453c952ae10bbee30fbb","subject":"Fix: optimize create volume for NFS","status":"MERGED","created":"2025-08-19 12:46:42.000000000","updated":"2025-09-15 20:11:01.000000000","submitted":"2025-09-15 20:09:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957891-new-location-api-support","meta_rev_id":"623f9e335269c9570828d7ace66f528e6e556e0a","_number":957891,"virtual_id_number":957891,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-15 20:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-09-12 13:58:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-09-15 17:39:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2025-09-05 19:27:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-09-05 19:27:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-09-15 17:39:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-09-12 13:58:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-19 12:46:54.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-19 12:47:40.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-08-19 12:51:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-08-19 13:44:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-19 13:47:36.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-19 13:56:47.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2025-08-19 14:54:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-19 15:07:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-19 15:11:52.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-19 18:19:43.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-09-04 18:14:04.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-05 12:36:30.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-12 13:58:08.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-09-15 17:39:10.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"4e238ba193bf6a1a7fdcb24ec9d4098ef3a1310a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-19 12:46:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01004efe6d000bac0f4cc8fde3032264c1673466","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-19 12:46:54.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/215/","accounts_in_message":[],"_revision_number":1},{"id":"1a211056f27d20089e411d32a0b48d0902cf8fbe","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-19 12:47:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"1d0476fb89221b05ae681d0ca2f0c63f2bdd9f36","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-19 12:51:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d478d8a29757d12a0d94506a21267dd6a87f436e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-19 13:00:21.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"03cdda48bf0b7abe64380ef75790e11c99bfd069","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 13:44:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19656f162bba492bbe91c485360e7e7a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d27e3f2a739f416dacaf90831e42914f : SUCCESS in 26m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a73902a661754a8ead90349b7562c4e8 : SUCCESS in 26m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/406b3c7407aa4fe8b34f5706d3418961 : SUCCESS in 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"386b7b8c3e876bdc52e9fb7d97ecec93539b468c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-19 13:47:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1480/.","accounts_in_message":[],"_revision_number":1},{"id":"27e8f0469dabf940ee5febfa4f4d35f6ac6b1598","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-19 13:48:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1615/.","accounts_in_message":[],"_revision_number":1},{"id":"7b2f54ae1031554cd6f1d1f1b4ecb27d0e29f18b","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-08-19 13:56:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a6232f36fae23b2b2cc176c820cea1dd7dda66b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 14:54:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1670e8e81514233b69feff7b35b8c77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8f32f53ce3e4cf1a2beccf76e2b7b63 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d659c77f7953495989b0cf3f9df51c7c : SUCCESS in 13m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f015cb6282c94fb9b644856fb1418107 : SUCCESS in 7m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9c0bd255ba149a79b3d4ed7198d1f9e : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dddf00d1e84d4401a14471a507ee811b : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/8374fb9616dc419fb6287438020ad75e : SUCCESS in 47m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ca87cdc120b4b44a41ed9ddcb854660 : SUCCESS in 32m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2042013db8c4790a6eaf2d93dcd1261 : SUCCESS in 1h 26m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/382c4bec85ee4f0c8e3977ba8b4c576e : SUCCESS in 31m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f1570bd3c094be5acee6ff01dc51dc8 : FAILURE in 5m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e3c64a47ee3c431f84295ba4dcbc7ff2 : SUCCESS in 22m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/be0fc600b19242de9fc1a4c3c8d04cf9 : SUCCESS in 2m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/325e985c253c4853822cb18b5baeedef : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4a2e5fff580d4c7c8eae32effc6429a3 : SUCCESS in 5m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c0e9e42f0a824a69a5a7577e6d7a5253 : SUCCESS in 7m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/19d55ad4ccdc40dea2d6d4ace4a5df6d : FAILURE in 18m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3237b4239df64926ad7105db84265444 : SUCCESS in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/317d5cf6fbea426c888edf4649daf494 : SUCCESS in 1h 59m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/499235c39035472e99ad371053d749a3 : FAILURE in 1h 47m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/370d71f0d395404ea39e960943484cda : SUCCESS in 1h 43m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3af6a88e983041e2920cb5bd51155862 : FAILURE in 13m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e2ed7323c85843b390c225608664dd86 : SUCCESS in 20m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/764af697283149f995972ebe23dbddc6 : SUCCESS in 1h 01m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5bd2dd6b511041a599694f036eb706b7 : SUCCESS in 1h 24m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7d095b09732a442c8f4e4db19cf70d45 : SUCCESS in 47m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/31afc195ac2548b3bdfa526fc1a1872d : SUCCESS in 57m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/edcafaaf3ad949ad93db55a39eef9812 : FAILURE in 12m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/baed60a6c5ba426d8c104e78f95e1a7d : SUCCESS in 1h 23m 03s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e8868d07d8a4d819e68a4938627c9b4 : SUCCESS in 1h 38m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/800d99cbc9f444ceadea6d5edb312a39 : SUCCESS in 37m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bbd9cb9f8960e10397648de75d0e512593c936a5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-19 15:07:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/12fb9e2661a44b71a0d897f84c80dd84 : SUCCESS in 49m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/66ecdf48fd1d42fe846a5858b62ccef5 : FAILURE in 55m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5ed6be9d91729045822c43b7f2d2dc443d91b80b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-19 15:11:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/957891 : FAILURE in 1h 24m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cac49076a0ea7a88a0de61f5f8b0550eb2bb07a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-19 16:22:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"de8dc60db21f5cdbae3d4aa64aa4678cd77a6488","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-19 16:23:03.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/219/","accounts_in_message":[],"_revision_number":2},{"id":"abf0bd5557bdb6ccc9653a6f85d560fae8c00266","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-19 16:25:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"90a6474649ca2e278cee08e1f7f3bfcb745ba569","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-19 16:37:07.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"f4b530811837165267c607a44fdc9676479b6029","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-08-19 17:02:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b5eb0347a9ce55c4c057a36240c3ef04ed083a75","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-19 18:19:43.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-91-957891-1/FC/ : FAILURE in 5h 30m 37s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-91-957891-1/iSCSI/ : FAILURE in 4h 9m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"ff2e48e90827d8a0a933a986b8d366567b4441bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 19:24:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3dc292a5ba34579bb41315314d55151\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4110623e96d3497eb9185ecf1fcd89e8 : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7baa33eec4aa4d57be665c2c8eb59e70 : SUCCESS in 24m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c5397a0123f842baba07369ed55a09ff : SUCCESS in 28m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21f4e11de1cf846754fc7d09d201fba1a370d361","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-19 19:46:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/957891 : FAILURE in 1h 23m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6d354efedaebe4787a9ef1dcf6549fa648dcf868","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-19 19:46:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/380c7051758a4bc9bc4d6d6ed982509d : SUCCESS in 45m 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e3a75878a3fd4eda995ec203b5d5db0a : FAILURE in 59m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3c7b6c608339d0b05db90b51e32cdaa61b561125","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-19 21:09:29.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-91-957891-2/FC/ : FAILURE in 4h 44m 53s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-91-957891-2/iSCSI/ : UNSTABLE in 2h 7m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"0447398ccca779aa5816e59d4939004fdb30deed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 21:18:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b032c8e529147e6a86443196e469d97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7c9d790a8b74e8cb6ebb11a42475f9b : SUCCESS in 8m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1a99e32a01147ce8a3ead96b52ef644 : SUCCESS in 9m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b27ad0423104c4db428e8ecc5819950 : SUCCESS in 10m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/948df6961c844f4aac911b74b80a0bcd : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb5d7bd432f64382bec356b4433d494b : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/11790acd321e4918a8d5874bd7d6e4ae : SUCCESS in 50m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/01564b56ad284cfb9f0f9356e285d930 : SUCCESS in 50m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/24a63425fc4e4cf3b09886809c510ae1 : SUCCESS in 1h 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44eaca981a6a4bb785e990c1b5336888 : SUCCESS in 1h 05m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c46973f24994249a41b65be7b1dcb40 : SUCCESS in 13m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/227aa91e89c147c1ad2a837a93b77335 : SUCCESS in 24m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ddaa113c518434ba33759ffc0091c4a : SUCCESS in 6m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/57851f3d85614b70becc35859d88d1e7 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b09da41b933b4616a2cf094e5838ad74 : SUCCESS in 7m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6c1b851d0d9347d69201b129b60c80f7 : SUCCESS in 10m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4d5d87ce0150419680fe1f9f1215a105 : FAILURE in 29m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c43e6c92ebdd4867b02cd77cfc8be76e : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/45b4a33b1e934f8095901f4d96d34cb8 : SUCCESS in 2h 01m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/36bd756a74bf4eb8aa4db6f77c31801c : TIMED_OUT in 3h 04m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea2fdd63beb54942a913fec03ae4e7da : SUCCESS in 1h 25m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/26b4c70619a943b8b7913d8dde669652 : FAILURE in 18m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8b529489d23447dc97a691106bdb605e : SUCCESS in 26m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/619854ec0bdd4b69a44082f08a5cb140 : SUCCESS in 1h 08m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/44ac4ef069134a07921987a82172b6e0 : SUCCESS in 1h 19m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/94fa22c1e3fd414c91f6a3cf1c626a3b : SUCCESS in 38m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/571756e55e6b43db9fda8ab6ae119d1a : SUCCESS in 46m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7a5d84cc2dd14ec08ff70d26d9e9a09e : FAILURE in 13m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c00b4a14ac8f486cb967b1f907e407b4 : SUCCESS in 1h 29m 18s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d00d9034f7746c0aa8af1d424cdcd51 : FAILURE in 1h 07m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ecb8772a6ad4b85b721a0096288dc77 : SUCCESS in 44m 26s","accounts_in_message":[],"_revision_number":2},{"id":"39895714cfaca15661196993934a5c357c520950","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-19 21:48:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"34f744323622b99f340a7aa54f99b2fac8b0a21c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 23:20:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/394dd3a9fadf4cefa552756155676859\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9f131471929c4b539480ac5de280751d : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f59bb86bea284b2e9b170c1177f6bf6e : SUCCESS in 26m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d9acda18a8724bf886f6869c3c3e4eb4 : SUCCESS in 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eae71bac5ac699fca1d98c44c71ecbb03e9efe70","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-19 23:31:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1487/.","accounts_in_message":[],"_revision_number":2},{"id":"1502739556c200cf5134ac19c513e5394b1af19e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-19 23:48:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1622/.","accounts_in_message":[],"_revision_number":2},{"id":"25ce2401965ecab136edc531890aa87a18eaf623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 00:41:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8a8654985cd40ac907387353039317b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7bae66dd9404623849ac194d8156438 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df93b1edad9d41d8bede29dd3f84259f : SUCCESS in 11m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc59e531553d493ca3550349328dbac4 : SUCCESS in 7m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2837e91f3c0f45f0b2f760b0c4a7c540 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e511d8bde1af4bf38746fc6c6d9af18a : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/32a7c972e711440088beae797ddace7f : SUCCESS in 1h 00m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/290d7222fa604d1f93bf67efd15c877f : SUCCESS in 1h 02m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/74839331a86a476fbbdd75bd10b14609 : SUCCESS in 1h 36m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bfda88fd92440ba855d0331908dc7d8 : SUCCESS in 58m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22114d42334f406a8c6938c78b76e27f : SUCCESS in 15m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d1eaca4b2f3f4937a0736674799e0106 : SUCCESS in 25m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c642eff3718f40da845873441605f8f3 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/45a1657ccebf461381a28dfbe507496c : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/958ec783e3054abbba905fed198eb2f6 : SUCCESS in 6m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7a4a79186aa04b839432dbe691b90b80 : SUCCESS in 10m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f36caa3ded9b4904a806b6ea051109d4 : FAILURE in 35m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d6be973f5b574a169f018da8134e441c : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/419422cce2494789b63f59020ffe13df : SUCCESS in 1h 53m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/677e44a3cfd24e019d74ae121feb8d45 : FAILURE in 2h 47m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5b347f3e41f944f9b5f21028e97355f6 : SUCCESS in 1h 39m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d90b24bef47d41039cbd599f44f461f1 : FAILURE in 20m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ec5f7fbd93aa4c209036ba72d8716dc5 : SUCCESS in 16m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b241dcfe9c99422eb74d2d7b4c91dc31 : SUCCESS in 35m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d8cf4f3c7d9a4a5f862ffaa0bfd4f6ed : SUCCESS in 1h 16m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/34982c8f67674ca580df958f21f40d1a : SUCCESS in 34m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9203d011505e43b8a5f2fb19e487f3de : SUCCESS in 1h 31m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dc146ca49dee48bb8174d446ffc49ab2 : FAILURE in 21m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d4c510d94cbe401b9a95cb69d7032802 : SUCCESS in 1h 42m 14s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d0ad60d46e9463c8c4ecb417e30757f : SUCCESS in 1h 22m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4ac78cb385241179d4a87c57b6a1d1c : SUCCESS in 38m 41s","accounts_in_message":[],"_revision_number":2},{"id":"da96364ca711f481f57ee5345bca17a60d0199b2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-20 00:46:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/37efd69f612d40d4a2896a2993f2f307 : SUCCESS in 43m 02s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b17b925b2eba48a9abc1761833d72666 : FAILURE in 55m 36s","accounts_in_message":[],"_revision_number":2},{"id":"52d380c5da9b3c08a9057b74ec6f56fb56c54208","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-08-20 06:18:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e1a668385382657276667a91608ff72d774f5b9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-04 18:11:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ccd63119304771acb4da99e28097e95708dece18","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-04 18:14:04.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"1a550b3ead24f60938018c5020b2db7bf94e1db5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-04 18:21:06.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"bfcddfbffe15eae08061b2daa51a40020f108199","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-04 18:21:41.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"ae8116bc2c6541cc794202ac9bbc3f391d17604b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-04 18:25:34.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"6a08874c1df4183fb1cbe530ded19fa9da2366ff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 20:46:19.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/457/","accounts_in_message":[],"_revision_number":3},{"id":"3cd3a968a3cdeeb71966ba18a2f3ee9fd9d8f73c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-04 20:56:38.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"4f48388b583e00409b36e47f8bcdd844497ba0e9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-04 20:57:09.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"4244ee813f4b749d328aac06a3180c4721a0f93b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-04 21:01:10.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"0d34e77847a9995aba56c27bfb5e9fd4a7bbdd22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 21:04:05.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/957589 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"280141b901b11d29e494e1c86dfaa7fcdb81d4a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 21:04:08.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/957589 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"fd5d8746d114d87994292d2c53fcc0165180b1d4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-04 21:04:26.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"d916111938285a597468c5c599588f9dd0609833","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 21:06:08.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"1de344874aaf68cbd64c90353e4e3353541438c8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-04 21:06:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f5ac2447b31f7c9d8b11ce94b4e325ba547b70e2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 21:44:41.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/460/","accounts_in_message":[],"_revision_number":4},{"id":"a24ff9ded9cce15e6b6874f6ece9386c1587c370","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 22:04:08.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"7f30d6905d2fbab53868ca246336a22cf6248c92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-04 22:11:32.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ddc04b99bd1165da5722d9a66659d67a1e2ee59b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-04 22:14:47.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"b08e900678403ce719665bb0acef7ff2e4f6f127","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 23:04:31.000000000","message":"Patch Set 5:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/464/","accounts_in_message":[],"_revision_number":5},{"id":"1b970f4b681f5a496c2c17af7d7afbb4b86fc04b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 23:09:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a2339c11d2240b5832bd129c491e329\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/de382f5ba9dd43a5a63deef0a02559eb : SUCCESS in 29m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0a3451449aae4a938dcf27dbd6a4e90b : SUCCESS in 29m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/757dabcb7f8b47ea902f501519c16cb6 : SUCCESS in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9a392e4e63a9cd075e3ea9f167644066b6be3a86","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 23:23:54.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"ccac905d690e5ca5edaf78ae0973af38f3e9ece8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 23:44:08.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/466/","accounts_in_message":[],"_revision_number":6},{"id":"e143a1bfb95bb08e6d55ca4e832c9b446c687bcd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-05 00:03:46.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"1cc3b4f264b60ae21b2583b0f5240975fb8ea2da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 00:48:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bee344cfba14abd814ecf0016f6c7bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/318a83130bc24bd2b17bffb0968743a1 : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59103b2e770b4840b319da669eec1a10 : SUCCESS in 11m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2861a15e938c4f64bc13f9455e884d29 : SUCCESS in 9m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19eba5b93e1a4b8d879c53df46be7014 : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38be18a311524b308885544d31dc3712 : SUCCESS in 15m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/ad28af74b2d6468bbe54f371f8da0fe6 : SUCCESS in 59m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c593c3e2a48e400cad02daa4c82d9ad7 : SUCCESS in 59m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c1c98328c6134ce693bbf549fb00d355 : SUCCESS in 1h 32m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0692f85b598d46599c25a38f43efc3c6 : SUCCESS in 1h 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d852613d10f2447dbb1807dc868ffa09 : SUCCESS in 15m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/424bb0ebaa6f45b6820307428d85fa0e : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/71a2f615d38348ad92836ed9c2e929e2 : SUCCESS in 8m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5c8a01cf0e10447fbaf1f771973c86f4 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b5cd64ec34794c37a68cbb678e472c2d : SUCCESS in 6m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5b7f45f02d824b2ea71783a54a5fb90f : SUCCESS in 9m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a7b6417b683444eab06b07145eff65a3 : FAILURE in 35m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3cedc60ee6ed42e689116949690fdced : SUCCESS in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd5e091f25b74784bb600eefd82c1aab : SUCCESS in 2h 07m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0630dcd5b0fe4430b06d5d06e9fac454 : FAILURE in 2h 27m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/35c167dc8b084aa2bf8fa5d9c2bd5b7e : SUCCESS in 2h 05m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/26a69110ed614277b45c7d3ba197664e : FAILURE in 18m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/769889154a0442ae8387b777319ef3f6 : SUCCESS in 33m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f3a5f715a8274043bea8390e24deebb5 : SUCCESS in 1h 10m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aafbec167c594505818c7c19eca8a020 : SUCCESS in 1h 41m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/527dbdc59402495e810cedf17c4a7604 : SUCCESS in 42m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/78f71ccdfcc049f698275f452fc95062 : SUCCESS in 1h 33m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/354a4a1ea8bc40a8955ea4879cfc793a : FAILURE in 20m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0085e01fd0ae41cca6c962b4c5da149b : SUCCESS in 1h 45m 58s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8e262dcc9ca437486de69f280cc91e9 : SUCCESS in 1h 39m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06b0ccc64ebf42388efdfab64e4b7733 : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":6},{"id":"1aaecdffb7525b5e68fec2d135ca89a3dbd78421","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-05 05:22:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/957891 : FAILURE in 1h 05m 08s","accounts_in_message":[],"_revision_number":6},{"id":"c02ff050392b6833b07095e8aab2a53d7e3234d1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-05 06:32:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1690/.","accounts_in_message":[],"_revision_number":6},{"id":"c30dd7be4b120337fc067a764b511908ea4a08a5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-05 06:44:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1826/.","accounts_in_message":[],"_revision_number":6},{"id":"32908468bd5536911ac041a52648e167381d2fb5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-05 09:07:46.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b075cd928a4e4e1ba54be01fd8ee102f : SUCCESS in 1h 32m 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e1df9e6313ca40e49fa265681bf94470 : RETRY_LIMIT in 3h 25m 54s","accounts_in_message":[],"_revision_number":6},{"id":"af7a189567c172ccce0dddd70f52ec189a26061a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-05 12:36:30.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":3},{"id":"1c5e3449ced031650d4ed4baf845dfb37bf067de","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-05 14:36:27.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":4},{"id":"bf57cf03f4f00fe489a93c4575e80b7928ba8c3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-05 19:27:23.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"874784a8c384693a6c71c3178fda62e9cc785dfa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-05 19:31:56.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"eab98123890f8979316b1056a11c3999d754ed14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 19:58:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/972834aa6b4141f48e307a9dcb3385dd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/caaf4c66e01f4ff2a3a4a00af0267aa5 : SUCCESS in 25m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5854d4f7500f4488ace0004be4468c40 : SUCCESS in 26m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/db0a4411c0cb417dbcb3bfd30df3d9ea : SUCCESS in 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cb7f5fb4acb6e0e686f1de4ccf307580b3f6fa09","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-05 20:10:25.000000000","message":"Patch Set 7:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/479/","accounts_in_message":[],"_revision_number":7},{"id":"ef03bf2f8c8229b3847412a37b5aee5178fb422a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-05 20:13:16.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":5},{"id":"e9a5cd91cc8162df35e5b97981886e8bf92ec8da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-05 20:29:18.000000000","message":"Patch Set 7:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957891/7/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"a232d9c523a0d3643feed22bf40bf6ba531b32a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 22:05:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/656744c9328e4a02a4ad9ad9afc211af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98fa8fc3cfe347ddb7ec5f0d79c9d1ce : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf386f6ec7924281bdb9f515b06cb655 : SUCCESS in 11m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2aa88e68cc54ae39a6bf0c6025afd51 : SUCCESS in 14m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/208aac2d854341f19a3cb6bcdd04f641 : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a67ad21a6b44e7a8b1d2ec2c16fe57e : SUCCESS in 13m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/baebe61344d74612beed89e6efaf900f : SUCCESS in 1h 12m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dfaee5d08c2a4e49b782706096d0d208 : SUCCESS in 1h 02m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/324f26f42b854a24a93aa36dca00a842 : SUCCESS in 1h 22m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ece2faba7634fa78000b9571558a8d0 : SUCCESS in 1h 06m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd45f080d3fe40b7bebc6568241b8031 : SUCCESS in 18m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bf90779314884df0834ba94177b4d5da : SUCCESS in 23m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f5dd137cb7c74f429dbca67d1e37ce25 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f26715cfc0c4739ac5085950f1ce4a7 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aa450fb509484afe8c1279592bceb49a : SUCCESS in 5m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b7f888b4de634d3a8b468f635130c025 : SUCCESS in 9m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5d9f13b83bbe459dae7b869543aa4496 : FAILURE in 41m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f606fe5e02b642aa82f66c87627492a7 : SUCCESS in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f05384fed8f3426d91ffb0e27fe0e911 : SUCCESS in 1h 30m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8238230a9dc5415a84318be320ab51b1 : FAILURE in 2h 32m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f83e2a14eb464fd595c55ada23cb2fb7 : SUCCESS in 1h 45m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/954b235d330b4506966ef1c244e5efc8 : FAILURE in 19m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b7125b4bf1f744459d32751411b883b1 : SUCCESS in 29m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/55bf1ed18688402a8563f9eded1b47eb : SUCCESS in 59m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/60cf0fc784554ff0a19113fab3c8efc0 : SUCCESS in 1h 46m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/48382bc784554e7eb250c26acd5d2b54 : SUCCESS in 49m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7a682b1366a34ddeac5351e3b35c6351 : SUCCESS in 1h 27m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/783bcad975cb4046b1a89418a469fcc6 : FAILURE in 12m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8066958360334ff292d28f1bebec1c63 : SUCCESS in 1h 22m 26s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc0337fe50134178b4c70e7119cc2d91 : SUCCESS in 1h 38m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c1bd295ae2f40edabd2e4991a73f4b9 : SUCCESS in 57m 51s","accounts_in_message":[],"_revision_number":7},{"id":"1faf11fdfb5b868b0e852a077e678961f6a585ad","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-05 22:37:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1838/.","accounts_in_message":[],"_revision_number":7},{"id":"ce9d2d48b9693b61929d61d0ea625869df7cc95c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-05 22:56:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1702/.","accounts_in_message":[],"_revision_number":7},{"id":"42551140aa5741e9d822507dfe9cfb7c0fa155a3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-05 23:06:54.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":6},{"id":"2813fdc8cd4f24618801e0697acbf4b66c419b4c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-06 02:23:29.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/957891 : FAILURE in 1h 53m 42s","accounts_in_message":[],"_revision_number":7},{"id":"67127e9917d0135c41f4c3b0219de8a0496109aa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-06 02:42:48.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/57d7b8c1880342da911e9209e701989f : SUCCESS in 1h 37m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ce12f73ba00942adb221e17d771daee6 : FAILURE in 1h 51m 10s","accounts_in_message":[],"_revision_number":7},{"id":"43bb135db8a188b26c4b296c88dd8c94b3b49409","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-06 08:49:15.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":7},{"id":"838e3e78fa4235b31705f354377668a340708379","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-09-12 13:58:08.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cabb673435cffc63ab39e55f07cc8117c6871a03","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-09-15 17:39:10.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"cc9fea4fd1fabce02f4160adaa259806d749d494","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 17:39:37.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2e019562a1308d08cb7c3577f133783d8e89726f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 20:09:48.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91eafb1be3a246eba7bd29bdf714ac36\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a7163eeb0d4c17b3ea5b7b5ada38b3 : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d8856650d784d1eafe97e31845894a5 : SUCCESS in 10m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c75097c231f4f15b1c2a5b86ba72d34 : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa2f09eb626440fb7e8bc8c20fdaa9d : SUCCESS in 14m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/c5e0ce8f5733446d8eb5090cd553f37a : SUCCESS in 44m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9683960306c424ca5231c2d23ff444b : SUCCESS in 55m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3475e8e2d23f47cf855727aeacf880a9 : SUCCESS in 1h 25m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74bf78772fa5430fbc5edfb0149c3d55 : SUCCESS in 1h 03m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fee8b4ed11e4f6fab9e96acd1f6789a : SUCCESS in 11m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2bf58c8ecd284d38a38fe964b772bc3f : SUCCESS in 58m 39s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/06bed1686dd14fd6aedf9b0d3b1f1ee8 : SUCCESS in 2h 19m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7120bb199b347be8ac30978bf903130 : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":7},{"id":"0450ff4cd6d4c285398e9d10e83362750dd24472","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 20:09:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"623f9e335269c9570828d7ace66f528e6e556e0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 20:11:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4bb802c4b9c457aa2487514a3b179c0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab49eb7a88d41e48f6c53ece191345f : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17258da82cbe483d81cc4f4911fa1ec7 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e564049d895d664f001ef51837987f5d06d6ae8b","revisions":{"11f9e67ffc94ef12e5568cbf8b8a69655d4e593b":{"kind":"REWORK","_number":1,"created":"2025-08-19 12:46:42.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/91/957891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/957891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/957891/1"}}},"commit":{"parents":[{"commit":"2dd403e19dd081ef997e103e91396e569c5e2811","subject":"Add support for GET location API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2dd403e19dd081ef997e103e91396e569c5e2811"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-19 12:20:18.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-19 12:20:18.000000000","tz":0},"subject":"Fix: optimize create volume for NFS","message":"Fix: optimize create volume for NFS\n\nWhen we enable optimized path for creating a bootable volume from\nimage, we call the driver\u0027s ``create_cloned_volume`` method to\nclone the image-volume to create the new volume.\n\nWhen Glance is using Cinder as a backend and Cinder is backed by\ngeneric NFS, we encounter the following error:\n\nAttributeError: \u0027Volume\u0027 object has no attribute \u0027obj_context\u0027\n\nAfter looking into the optimization code path, we can see that the\n``create_cloned_volume`` method is called with SQLAlchemy object[1][2]\ninstead of the oslo versioned volume object. When the remotefs\ndriver tries to fetch the context from it (obj_context), we fail.\nThis is not faced by the LVM or RBD driver as they only access\nthe properties that are available in the SQLAlchemy object.\n\nThe right thing to do here is to pass the OVO object to the drivers\nwhich this patch intends to do.\nNOTE: This issue is not specific to the NFS driver but since that\noperation is broken, the releasenote is intentionally written that\nway to aware operators that the issue is fixed.\n\n[1] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L750-L755\n[2] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L781\n\nCloses-Bug: #2120933\nChange-Id: Ibfbcff01c761a569587e453c952ae10bbee30fbb\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11f9e67ffc94ef12e5568cbf8b8a69655d4e593b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11f9e67ffc94ef12e5568cbf8b8a69655d4e593b"}]},"branch":"refs/heads/master"},"4fdcea5d4a89ea65e8d33ebae152775cc225bb4e":{"kind":"REWORK","_number":2,"created":"2025-08-19 16:22:55.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/91/957891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/957891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/957891/2"}}},"commit":{"parents":[{"commit":"2dd403e19dd081ef997e103e91396e569c5e2811","subject":"Add support for GET location API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2dd403e19dd081ef997e103e91396e569c5e2811"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-19 12:20:18.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-19 16:20:38.000000000","tz":0},"subject":"Fix: optimize create volume for NFS","message":"Fix: optimize create volume for NFS\n\nWhen we enable optimized path for creating a bootable volume from\nimage, we call the driver\u0027s ``create_cloned_volume`` method to\nclone the image-volume to create the new volume.\n\nWhen Glance is using Cinder as a backend and Cinder is backed by\ngeneric NFS, we encounter the following error:\n\nAttributeError: \u0027Volume\u0027 object has no attribute \u0027obj_context\u0027\n\nAfter looking into the optimization code path, we can see that the\n``create_cloned_volume`` method is called with SQLAlchemy object[1][2]\ninstead of the oslo versioned volume object. When the remotefs\ndriver tries to fetch the context from it (obj_context), we fail.\nThis is not faced by the LVM or RBD driver as they only access\nthe properties that are available in the SQLAlchemy object.\n\nThe right thing to do here is to pass the OVO object to the drivers\nwhich this patch intends to do.\nNOTE: This issue is not specific to the NFS driver but since that\noperation is broken, the releasenote is intentionally written that\nway to aware operators that the issue is fixed.\n\n[1] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L750-L755\n[2] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L781\n\nCloses-Bug: #2120933\nChange-Id: Ibfbcff01c761a569587e453c952ae10bbee30fbb\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4fdcea5d4a89ea65e8d33ebae152775cc225bb4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4fdcea5d4a89ea65e8d33ebae152775cc225bb4e"}]},"branch":"refs/heads/master"},"fdf33dee5078bf8e51969410e597cd80b192a2dd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-04 18:11:23.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/91/957891/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/957891/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/957891/3"}}},"commit":{"parents":[{"commit":"696b9b7afa674bb8303a4c1d3f760a2674c1d8fe","subject":"Add support for GET location API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/696b9b7afa674bb8303a4c1d3f760a2674c1d8fe"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-19 12:20:18.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-09-04 18:11:23.000000000","tz":0},"subject":"Fix: optimize create volume for NFS","message":"Fix: optimize create volume for NFS\n\nWhen we enable optimized path for creating a bootable volume from\nimage, we call the driver\u0027s ``create_cloned_volume`` method to\nclone the image-volume to create the new volume.\n\nWhen Glance is using Cinder as a backend and Cinder is backed by\ngeneric NFS, we encounter the following error:\n\nAttributeError: \u0027Volume\u0027 object has no attribute \u0027obj_context\u0027\n\nAfter looking into the optimization code path, we can see that the\n``create_cloned_volume`` method is called with SQLAlchemy object[1][2]\ninstead of the oslo versioned volume object. When the remotefs\ndriver tries to fetch the context from it (obj_context), we fail.\nThis is not faced by the LVM or RBD driver as they only access\nthe properties that are available in the SQLAlchemy object.\n\nThe right thing to do here is to pass the OVO object to the drivers\nwhich this patch intends to do.\nNOTE: This issue is not specific to the NFS driver but since that\noperation is broken, the releasenote is intentionally written that\nway to aware operators that the issue is fixed.\n\n[1] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L750-L755\n[2] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L781\n\nCloses-Bug: #2120933\nChange-Id: Ibfbcff01c761a569587e453c952ae10bbee30fbb\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fdf33dee5078bf8e51969410e597cd80b192a2dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fdf33dee5078bf8e51969410e597cd80b192a2dd"}]},"branch":"refs/heads/master","description":"Rebase"},"f94dc55119303781743d4683e95634886bc982ec":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-09-04 18:21:06.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/91/957891/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/957891/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/957891/4"}}},"commit":{"parents":[{"commit":"a953f77412e1341591bbd0848be674ac2664c325","subject":"Add support for GET location API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a953f77412e1341591bbd0848be674ac2664c325"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-19 12:20:18.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-09-04 18:20:06.000000000","tz":0},"subject":"Fix: optimize create volume for NFS","message":"Fix: optimize create volume for NFS\n\nWhen we enable optimized path for creating a bootable volume from\nimage, we call the driver\u0027s ``create_cloned_volume`` method to\nclone the image-volume to create the new volume.\n\nWhen Glance is using Cinder as a backend and Cinder is backed by\ngeneric NFS, we encounter the following error:\n\nAttributeError: \u0027Volume\u0027 object has no attribute \u0027obj_context\u0027\n\nAfter looking into the optimization code path, we can see that the\n``create_cloned_volume`` method is called with SQLAlchemy object[1][2]\ninstead of the oslo versioned volume object. When the remotefs\ndriver tries to fetch the context from it (obj_context), we fail.\nThis is not faced by the LVM or RBD driver as they only access\nthe properties that are available in the SQLAlchemy object.\n\nThe right thing to do here is to pass the OVO object to the drivers\nwhich this patch intends to do.\nNOTE: This issue is not specific to the NFS driver but since that\noperation is broken, the releasenote is intentionally written that\nway to aware operators that the issue is fixed.\n\n[1] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L750-L755\n[2] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L781\n\nCloses-Bug: #2120933\nChange-Id: Ibfbcff01c761a569587e453c952ae10bbee30fbb\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f94dc55119303781743d4683e95634886bc982ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f94dc55119303781743d4683e95634886bc982ec"}]},"branch":"refs/heads/master"},"d5878e8c47c6deabeaa5792b6b6a34369e893783":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-09-04 20:56:38.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/91/957891/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/957891/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/957891/5"}}},"commit":{"parents":[{"commit":"ac0c4c0a8d96625e4cd4d7c4eed1d0ba953b2fb8","subject":"Add support for GET location API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac0c4c0a8d96625e4cd4d7c4eed1d0ba953b2fb8"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-19 12:20:18.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-09-04 20:55:56.000000000","tz":0},"subject":"Fix: optimize create volume for NFS","message":"Fix: optimize create volume for NFS\n\nWhen we enable optimized path for creating a bootable volume from\nimage, we call the driver\u0027s ``create_cloned_volume`` method to\nclone the image-volume to create the new volume.\n\nWhen Glance is using Cinder as a backend and Cinder is backed by\ngeneric NFS, we encounter the following error:\n\nAttributeError: \u0027Volume\u0027 object has no attribute \u0027obj_context\u0027\n\nAfter looking into the optimization code path, we can see that the\n``create_cloned_volume`` method is called with SQLAlchemy object[1][2]\ninstead of the oslo versioned volume object. When the remotefs\ndriver tries to fetch the context from it (obj_context), we fail.\nThis is not faced by the LVM or RBD driver as they only access\nthe properties that are available in the SQLAlchemy object.\n\nThe right thing to do here is to pass the OVO object to the drivers\nwhich this patch intends to do.\nNOTE: This issue is not specific to the NFS driver but since that\noperation is broken, the releasenote is intentionally written that\nway to aware operators that the issue is fixed.\n\n[1] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L750-L755\n[2] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L781\n\nCloses-Bug: #2120933\nChange-Id: Ibfbcff01c761a569587e453c952ae10bbee30fbb\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5878e8c47c6deabeaa5792b6b6a34369e893783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5878e8c47c6deabeaa5792b6b6a34369e893783"}]},"branch":"refs/heads/master"},"3f4632f6895e6547432b1053d84f3400160650a8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-09-04 22:11:32.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/91/957891/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/957891/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/957891/6"}}},"commit":{"parents":[{"commit":"0bc5244d46672015d5545112280365d76f94cd90","subject":"Add support for GET location API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0bc5244d46672015d5545112280365d76f94cd90"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-19 12:20:18.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-09-04 22:10:37.000000000","tz":0},"subject":"Fix: optimize create volume for NFS","message":"Fix: optimize create volume for NFS\n\nWhen we enable optimized path for creating a bootable volume from\nimage, we call the driver\u0027s ``create_cloned_volume`` method to\nclone the image-volume to create the new volume.\n\nWhen Glance is using Cinder as a backend and Cinder is backed by\ngeneric NFS, we encounter the following error:\n\nAttributeError: \u0027Volume\u0027 object has no attribute \u0027obj_context\u0027\n\nAfter looking into the optimization code path, we can see that the\n``create_cloned_volume`` method is called with SQLAlchemy object[1][2]\ninstead of the oslo versioned volume object. When the remotefs\ndriver tries to fetch the context from it (obj_context), we fail.\nThis is not faced by the LVM or RBD driver as they only access\nthe properties that are available in the SQLAlchemy object.\n\nThe right thing to do here is to pass the OVO object to the drivers\nwhich this patch intends to do.\nNOTE: This issue is not specific to the NFS driver but since that\noperation is broken, the releasenote is intentionally written that\nway to aware operators that the issue is fixed.\n\n[1] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L750-L755\n[2] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L781\n\nCloses-Bug: #2120933\nChange-Id: Ibfbcff01c761a569587e453c952ae10bbee30fbb\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f4632f6895e6547432b1053d84f3400160650a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f4632f6895e6547432b1053d84f3400160650a8"}]},"branch":"refs/heads/master"},"e564049d895d664f001ef51837987f5d06d6ae8b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-09-05 19:27:23.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/91/957891/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/957891/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/957891/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/957891/7"}}},"commit":{"parents":[{"commit":"b851d08e875bb734e3c1b43732ad5490f7b44f9e","subject":"Add support for GET location API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b851d08e875bb734e3c1b43732ad5490f7b44f9e"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-08-19 12:20:18.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-09-05 19:26:37.000000000","tz":0},"subject":"Fix: optimize create volume for NFS","message":"Fix: optimize create volume for NFS\n\nWhen we enable optimized path for creating a bootable volume from\nimage, we call the driver\u0027s ``create_cloned_volume`` method to\nclone the image-volume to create the new volume.\n\nWhen Glance is using Cinder as a backend and Cinder is backed by\ngeneric NFS, we encounter the following error:\n\nAttributeError: \u0027Volume\u0027 object has no attribute \u0027obj_context\u0027\n\nAfter looking into the optimization code path, we can see that the\n``create_cloned_volume`` method is called with SQLAlchemy object[1][2]\ninstead of the oslo versioned volume object. When the remotefs\ndriver tries to fetch the context from it (obj_context), we fail.\nThis is not faced by the LVM or RBD driver as they only access\nthe properties that are available in the SQLAlchemy object.\n\nThe right thing to do here is to pass the OVO object to the drivers\nwhich this patch intends to do.\nNOTE: This issue is not specific to the NFS driver but since that\noperation is broken, the releasenote is intentionally written that\nway to aware operators that the issue is fixed.\n\n[1] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L750-L755\n[2] https://opendev.org/openstack/cinder/src/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b/cinder/volume/flows/manager/create_volume.py#L781\n\nCloses-Bug: #2120933\nChange-Id: Ibfbcff01c761a569587e453c952ae10bbee30fbb\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e564049d895d664f001ef51837987f5d06d6ae8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e564049d895d664f001ef51837987f5d06d6ae8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
