)]}'
{"id":"openstack%2Fcinder~957942","triplet_id":"openstack%2Fcinder~master~I670b85531576190daa5834da693bb5da687f30b9","project":"openstack/cinder","branch":"master","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-08-19 15:26:33.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I670b85531576190daa5834da693bb5da687f30b9","subject":"Add \"mypy-head\" tox environment","status":"NEW","created":"2025-08-19 15:26:24.000000000","updated":"2025-08-28 09:23:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c7fc13119d7abc3ceb44de5730cc0c531097c74d","_number":957942,"virtual_id_number":957942,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-19 18:23:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-19 15:26:33.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-19 18:23:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-19 18:45:22.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-19 22:33:51.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-28 09:23:14.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"e4cdb50ccf508bdf764a003ebcf3f2a46ca72a0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-08-19 15:26:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"193c32f06d1a188073520de4a90bb32699174706","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-19 15:26:33.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/218/","accounts_in_message":[],"_revision_number":1},{"id":"26852c179f2bfab425136ed34240ab4c547bda85","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-19 15:40:20.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957942/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/957942/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"5efecd434441870e5c084dc639b041f866682001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 18:23:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fbac833b1184eda94e39ae3531b350d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bf2ef3166cbd497a87569641eac4dbbc : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b037b84825344ca5a250b14580c533cf : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15f801c11682487987e4ea457ebc609b : SUCCESS in 10m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46bae9fb3639485ea80d898de882ee22 : SUCCESS in 10m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2e328cc2a6d4dae8f3e8b94b24c565b : SUCCESS in 13m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d1e857006147958c5da46a36927ed5 : SUCCESS in 7m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bdad41660d234ad6bf89c7a6fdad25ca : SUCCESS in 48m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44adaa618d96400dbed90882b77cdfef : SUCCESS in 8m 59s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/267cd1a49eff4d09ac541bb42ab8e55c : SUCCESS in 7m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6f7cd4fa3a9941a78adfdf934f172ca2 : SUCCESS in 5m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8e40d14e71c846ffbf53204b59b0ac30 : SUCCESS in 11m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1cbeb74d7c914a3d81d4df58288dd085 : FAILURE in 38m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/19e9aeb6ecfd4ff2a755d147f8034af5 : SUCCESS in 6m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3763683080ac401ea87b1196d3a145f0 : SUCCESS in 1h 34m 08s","accounts_in_message":[],"_revision_number":1},{"id":"db86d3ad863f67cece4198635105c15034c20e62","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-19 18:45:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5bc3cb0aa1a64d0aa159677cc90fc778 : SUCCESS in 48m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e95c6dc4d8674fd484f2720b5fe08358 : FAILURE in 1h 03m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0cfb9ef7f75b43faa1f3d49345db771f3f058d4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 19:16:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff12e9d7ded24e7db1f3e2f4d29f2737\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/697ec1a6402744d28935efb0fb4f1beb : SUCCESS in 23m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7ac584ad613f49969d9a1741d5ff5f64 : SUCCESS in 26m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/397b3729a8a24c8ab96c8b174a970648 : SUCCESS in 29m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ab513720e411cb629e0b301152dae178d5128ea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-19 22:33:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1486/.","accounts_in_message":[],"_revision_number":1},{"id":"868e85076306dcacbfaf6ef2255621e397afe98c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-19 22:39:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1621/.","accounts_in_message":[],"_revision_number":1},{"id":"c7fc13119d7abc3ceb44de5730cc0c531097c74d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-08-28 09:23:14.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/42/957942/1/20250828075558/NEC-ISCSI : FAILURE in 10h 13m 27s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/42/957942/1/20250828075558/NEC-FC : FAILURE in 10h 27m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58d23cf8ba174afa119896de9a13492f39034386","revisions":{"58d23cf8ba174afa119896de9a13492f39034386":{"kind":"REWORK","_number":1,"created":"2025-08-19 15:26:24.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/42/957942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/957942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/957942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/957942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/957942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/957942/1"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-08-19 15:23:53.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-08-19 15:24:40.000000000","tz":-240},"subject":"Add \"mypy-head\" tox environment","message":"Add \"mypy-head\" tox environment\n\nThis pulls the latest un-released mypy code and runs\nmypy checks with it. This allows us to detect problems\nthat will appear upon then next release of mypy.\n\nChange-Id: I670b85531576190daa5834da693bb5da687f30b9\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58d23cf8ba174afa119896de9a13492f39034386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58d23cf8ba174afa119896de9a13492f39034386"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
