)]}'
{"id":"openstack%2Fcinder~958465","triplet_id":"openstack%2Fcinder~master~I991e7481b0a9e031c303d60c5a5ab77fc257341c","project":"openstack/cinder","branch":"master","topic":"pbr","attention_set":{},"removed_from_attention_set":{"37328":{"account":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"last_update":"2026-05-11 06:56:09.000000000","reason":"Change was abandoned"},"36800":{"account":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"last_update":"2026-04-13 08:08:39.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2026-01-14 08:18:23.000000000","reason":"removed on reply"},"38067":{"account":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"last_update":"2026-03-12 06:01:32.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"34420":{"account":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"last_update":"2026-02-05 13:04:32.000000000","reason":"removed on reply"},"35429":{"account":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"last_update":"2026-05-11 06:56:09.000000000","reason":"Change was abandoned"},"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-09-04 04:27:05.000000000","reason":"removed on reply"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-05-11 06:56:09.000000000","reason":"Change was abandoned"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-02-10 18:42:21.000000000","reason":"\u003cGERRIT_ACCOUNT_10058\u003e replied on the change","reason_account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},"37470":{"account":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"last_update":"2026-05-11 06:56:09.000000000","reason":"Change was abandoned"},"38702":{"account":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"last_update":"2026-05-11 06:56:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I991e7481b0a9e031c303d60c5a5ab77fc257341c","subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","status":"ABANDONED","created":"2025-08-25 18:10:54.000000000","updated":"2026-06-19 20:01:11.000000000","total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"41bc7df7d0e092d3cf84dd8e822daf541937a3a3","_number":958465,"virtual_id_number":958465,"owner":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-30 18:03:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2026-04-30 15:08:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":1,"date":"2026-04-30 15:08:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":1,"date":"2026-04-30 15:08:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-25 18:13:49.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"state":"REVIEWER"},{"updated":"2025-08-25 18:15:33.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-08-25 18:32:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 01:20:50.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-26 03:39:21.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-26 03:49:31.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-08-26 04:08:33.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-26 22:03:34.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-02 12:29:13.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2025-09-03 06:22:12.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-03 15:17:08.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-08 16:57:22.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 19:11:31.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-11-12 10:23:35.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"CC"},{"updated":"2025-11-12 10:23:44.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REMOVED"},{"updated":"2025-11-20 08:09:44.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2026-01-05 17:34:21.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-07 19:09:31.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-01-08 05:38:29.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-01-08 06:24:34.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-01-14 08:18:21.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-16 14:25:33.000000000","updated_by":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"reviewer":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"state":"CC"},{"updated":"2026-01-16 15:18:41.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-02-06 15:13:45.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REVIEWER"},{"updated":"2026-03-06 18:06:34.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-03-12 08:13:44.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-04-28 20:27:10.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-04-28 20:31:10.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"}],"messages":[{"id":"5d42fc2a193e920c5e895427adec62d2e4e22bd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-08-25 18:10:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f3a5e423b4f42b9b526aaa883593840dd5eca16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-08-25 18:12:13.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"133a5c3c9d9aafc129b62719f0ece2ac366cdad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-08-25 18:13:12.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"151b943a4a3272f23b71d6986c0fbe4e6ea4db50","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-25 18:15:33.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"eec5374ea8567d6a3281c08b11d9af5767f0ccb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 18:32:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e396327ed2b439b9ceb88a84e06b84e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c349f7cc27f4289bcfdca18e88aeaeb : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a4e4dece75344619642311c1d4a0d75 : SUCCESS in 14m 28s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f54adb2596ec479c82e20be293720f63 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"256ece676a32cf2077d0c0ec726a52ada1d609e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 01:20:50.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/294/","accounts_in_message":[],"_revision_number":1},{"id":"80ba0e85662d6af8095e23e6c161695ac8ccc7f3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 02:15:36.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"7eaabcfe6349a7a3b992c5de89405feea3271fe6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 02:15:37.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/295/","accounts_in_message":[],"_revision_number":2},{"id":"c44b775911f4b6e0af637256a7d721bdca450c3f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 03:11:55.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"42e2e4d6d4361658c93346223a9e3ae84e5bbd20","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 03:11:56.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/296/","accounts_in_message":[],"_revision_number":3},{"id":"31980f5bb5f83435996e3e0bafc16ef860055cbe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 03:39:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1695/.","accounts_in_message":[],"_revision_number":3},{"id":"eae7e647b50c300d24a5ae97306698f1ede79c77","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-26 03:49:31.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-958465-1/FC/ : FAILURE in 9h 36m 44s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-958465-1/iSCSI/ : FAILURE in 9h 9m 24s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"1637280fa4eff1ec90fdbda5bda668957f692841","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 04:08:32.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"13aba92f1a5d66f680e8d2dea5c8c4fd13954e1d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-26 04:08:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 1h 25m 06s","accounts_in_message":[],"_revision_number":3},{"id":"0214ee6a0f258a8a7b22a7fc5108678695611928","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 06:12:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1561/.","accounts_in_message":[],"_revision_number":3},{"id":"0c5e29a3880c622f04f43ed1be7cb63d6204f051","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-26 09:07:34.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-958465-2/FC/ : FAILURE in 12h 15m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-958465-2/iSCSI/ : UNSTABLE in 9h 48m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"67ab8536dcacaacb75e7340a73c4b99da82e86a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-08-26 10:48:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"53c19303f934f9faffaa6162b5796a67030c56b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-08-26 10:51:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ad33110c9bdd2d1626abfe295e77b184ec42de96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 11:05:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4af8559736fd44a0aba9dbfed4de3e03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3558ec6da45b4c0493f8e11a079e201b : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0179f7d2ca2840c3907967dcb91ba468 : SUCCESS in 12m 31s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/425a870b653b4567bd1789167b13825f : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":5},{"id":"600408a76732935903e9a558c0ccd62ceffe3b17","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 11:24:10.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/305/","accounts_in_message":[],"_revision_number":4},{"id":"30cadee4adfe064ecb68a689e023ae9cd88fdf1c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-26 11:33:17.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 36m 58s","accounts_in_message":[],"_revision_number":5},{"id":"bed6d0a7305c5a1f9ba83d90317f8b9173d0afc8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 11:53:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1702/.","accounts_in_message":[],"_revision_number":5},{"id":"cb97e13b7b246728d4958bd865a6c0cccc9aa65e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 12:17:45.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"f555b498c9cf25cc5babb7718b6c547fdcfcec15","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 12:17:47.000000000","message":"Patch Set 5:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/306/","accounts_in_message":[],"_revision_number":5},{"id":"d4051e50650ef37f8a7fa4a02941a537e5c6fe73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 13:11:24.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"07fe346b9eebb42b155c1c03b93a801473f04210","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 13:53:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1567/.","accounts_in_message":[],"_revision_number":5},{"id":"c4aa30094ac68b953fad596aa9768e9e1d1e128b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-26 15:46:21.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-958465-5/FC/ : UNSTABLE in 28m 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-958465-5/iSCSI/ : FAILURE in 4h 2m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"f29673f6843ac0a841837ad0f008b11ce2196727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-08-26 17:46:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e7398a75560145d8933799dbf28667fedd95b190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 18:05:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/051914934c6b48e1bcdca45333880294\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a87ab0d5ba024b60b7f419df9a2401b5 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a528e8acd6a740a681b0229ab2f68d82 : SUCCESS in 14m 24s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/69e58e9e51af409ebf490ea4961a2d5e : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":6},{"id":"3af3bcf115bb8bd4f600ab2e00caaa1fd582f87b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-26 19:50:11.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 36m 43s","accounts_in_message":[],"_revision_number":6},{"id":"223a05156166a3ea4681d74b6097d1f1d2974fe0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 20:02:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1573/.","accounts_in_message":[],"_revision_number":6},{"id":"1fdbbb51b9c75522179efacc083381694ac6ac9e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 20:09:24.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/312/","accounts_in_message":[],"_revision_number":6},{"id":"2aad3f236bd7ff11da60c320f24114f189565644","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 20:13:37.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1708/.","accounts_in_message":[],"_revision_number":6},{"id":"6e0628695b75ffbfbe149a8e9379a38efcb87a8e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 21:04:41.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"1a386a0b3d7cd90ac518973438a1961c622b9366","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-26 22:03:34.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"75eaddda29c066a98505d786b4d856c4ea43154e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-26 23:06:53.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-958465-6/FC/ : FAILURE in 5h 19m 35s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-958465-6/iSCSI/ : FAILURE in 4h 1m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"18429a5c534faef0970c35525f93b238a3151631","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-01 06:51:59.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"8d75102fdb54a45317fcbb06e84213c703ee5c09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-01 06:52:40.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"34a94fc867c4b583ff160be4fc1d089bc177f1a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-01 06:53:38.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"cad3559f45d204abd3aac9aa74698b575095ca5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 07:04:55.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/406dd626be5447a988a77d9d8272844a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a519f0e7046d4f9ba4b8b050c01a963f : FAILURE in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e533d921084830bf529cc1e4b6fea4 : FAILURE in 6m 19s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d2ca3e3b3d0b44b496b7c24efdd2f6a8 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":9},{"id":"785fac2b8ea48e2b9da8c377290b96493b738936","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-01 07:16:01.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"f3dd6a12ccffbc598c2092c975f1c31e825bb101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-01 07:16:38.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"20f17012b11c511e79094abb0e91c6a28d571c31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 07:35:10.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e05fe3d2cdb436b8a1bb44ba82c6537\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4187e0c5a7040bfad1f2a85834233dc : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e06aa6f4534fa8a9d4e17aa5d2099b : SUCCESS in 13m 53s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4b8735d954c045eb8c792a7c6ec994b7 : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":11},{"id":"8c65e182f1cbef04d7401c0b40f5d93ae1c15d27","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 08:51:17.000000000","message":"Patch Set 7:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/381/","accounts_in_message":[],"_revision_number":7},{"id":"81938d5c241d6361a8e2da85c4332755d5df6ea2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-01 09:17:47.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1768/.","accounts_in_message":[],"_revision_number":11},{"id":"a5b0bc072720b8e6f7806fa2965cae2cbf70c8f3","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-09-01 09:43:16.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"679b9211f26b51aa9b935c9f6dcae485a6fe23fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 09:47:40.000000000","message":"Patch Set 7:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/7/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"46b7ad8ed27c0d397792b1c0bd2799d59b74892a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 09:47:42.000000000","message":"Patch Set 8:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/382/","accounts_in_message":[],"_revision_number":8},{"id":"0adf5694cbae8f168d81dc0cb021eb5328611783","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-01 09:54:34.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1634/.","accounts_in_message":[],"_revision_number":11},{"id":"97a550e766923fbd36b93948b9a863df01abb1a2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-01 09:59:50.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"57aa307945a4a7e1a919e4eb96fd253a4ca72ffd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-01 10:31:50.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 1h 25m 45s","accounts_in_message":[],"_revision_number":11},{"id":"2c99b1331990b6093325f54d4c2621e38b42a576","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 10:44:51.000000000","message":"Patch Set 8:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/8/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"f0726a5f8ebf8a7682e9c3575a119581ec797082","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 10:44:53.000000000","message":"Patch Set 9:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/383/","accounts_in_message":[],"_revision_number":9},{"id":"8d42bcc543aca8ae132c2d1ab17c99d8d32c5360","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-01 11:30:57.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"9f7cd0e4c3c32c20b388c588858d1a535888beb8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 11:41:07.000000000","message":"Patch Set 9:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/9/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"6aa941413e6fdf79b14ee8b292cc0a951c479c35","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 11:41:08.000000000","message":"Patch Set 10:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/384/","accounts_in_message":[],"_revision_number":10},{"id":"574c9d3ff5e1a3af464e5035e55ba3448ed199b9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-01 11:49:28.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"cec92435c089fecfb38244f75476bb47799785f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 12:38:00.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/10/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"c755b91a38b8598c7c5efe0bdf87803400e53f35","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 12:38:02.000000000","message":"Patch Set 11:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/385/","accounts_in_message":[],"_revision_number":11},{"id":"2df5b240eb9ef050d8f366e55ebae71b96f210e8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-01 13:30:06.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"4b7b8bae7ade14e1356f55c41f8d0506f1fc224c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 13:34:30.000000000","message":"Patch Set 11:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/11/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"2941f2aa2594438c3035a96b03f1820f2841167f","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-09-02 12:29:13.000000000","message":"Patch Set 11: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"0b5ae217778b7c1eb7dd9722d38fe1b773c44c3b","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-09-03 05:44:03.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"e27893f24d83c9a375473cbb059e724130a9afe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-09-03 05:59:35.000000000","message":"Patch Set 12: Patch Set 11 was rebased on behalf of \u003cGERRIT_ACCOUNT_37470\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"_revision_number":12},{"id":"baa78cb69d46c83f9713f96b51c86c2fdd1788f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-03 06:11:02.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":13},{"id":"9757adc69f64621b60bf605a4bbc50812eb139c8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-03 06:22:12.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"a4bb0bdefeaf9d4759bfaa19281a0f7e61a4f194","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-03 06:56:15.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"4d7fd6e9d54937168a50fec89728914abab2f9a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-03 06:57:51.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"7c46dde490b35e95bff1792cc45f473ac3e752e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-03 06:58:32.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"4470e737e345448e855f1bc1de6bde150a09decd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-03 06:59:27.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"7a920cb5f03dc8d0c34b4cd234fd2a9419828f05","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-03 07:02:15.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":17},{"id":"25a761b898e5c7ec98187645a1aefdfd5fa8e5b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 07:29:19.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb8116885b60471da6d2dfd4fbf44fe7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c404786132df45ba9db3ab3002092567 : SUCCESS in 20m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1742e083cc8c4353816de685ca6627e4 : SUCCESS in 22m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/55f3a5d79c394bbbb76631fafc35bdd8 : SUCCESS in 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ad9044c8f0dad3543c4faf5e044b1c92ed9168a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-03 07:32:35.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"5a2e1521ffa7f42d9449c8cb289df467af34d3a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-03 07:33:04.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"03a0e7f4f785727ac9314359999bb76308a26c37","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-03 07:36:21.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":19},{"id":"ef9f89be30349d791481dc2ad25ff9afe49830f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 08:03:38.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ffa389f30fd42649733c4fdd37fbcb8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f148b82e4cdb428893c8860d0377ef80 : SUCCESS in 17m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66fe3089eac94c0eb6620046b48a1e39 : SUCCESS in 22m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/be9ee6953fe6497ab5d8f5088c8441b5 : SUCCESS in 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"ab21cfbf17d7e735ad04394536e5adc9ee5cabbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 10:33:52.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2eb0727e7ccf4b42a4c85c63063b8715\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1d1a5021c8b461db00de57ff57deb49 : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/891ee02d35114e5680da02bf0cb77eb3 : SUCCESS in 11m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1b47773c32f4611a6d1d6ab4c258e9b : SUCCESS in 9m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1041f691f0944549b5fd93a929d1dabb : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8642d183aa4e1085ca706fb21b8a04 : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/c54afc2bdc6e47afa89383246ce52488 : SUCCESS in 28m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/515fc2253794410abe713f59798d9ada : SUCCESS in 56m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b1dc17d883f4d3da3bc819a32bbdf5c : SUCCESS in 1h 21m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a13f23f2ea1430ebd419445f945db45 : SUCCESS in 58m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/52e0140adf6f48068ec38cb8cdccea60 : TIMED_OUT in 41m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/deb0499f65124382bc07265a9759366a : SUCCESS in 5m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f3d36fcc3653410095087fa1f1a40f7d : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bad62824d6644bab97c70c16ee7f56a1 : SUCCESS in 4m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6720d91ed868438f9ffd9f3f49203281 : SUCCESS in 5m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5116bbe366a24bef9f0b2ee4f0870f0a : FAILURE in 35m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5dd4d551253642d5873b90cfaf522eeb : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b2a3564c3f5c4767a783b0684afa0c07 : SUCCESS in 1h 53m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c9ec6dd7a5f349cd9405e210418560b6 : FAILURE in 2h 54m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a661ee94d94340bd93080b9a8c5aaa8a : SUCCESS in 1h 53m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ea5cbcaae7334079a3985c62a6194e7e : FAILURE in 18m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ea8462ea6e2d496795ffc902e6bbd9d2 : SUCCESS in 35m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d052b642239341368fbe9d089d341a5d : SUCCESS in 35m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9df51433349942acaa52be8603fdf6e0 : SUCCESS in 1h 36m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8c4e10d07500484eb618a0a4b0e69683 : SUCCESS in 46m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3e37876938124433961e1cf2fc000a2a : POST_FAILURE in 1h 48m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8fbf475c899a4bf9afaf6337d3e0ebb7 : FAILURE in 17m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ac51793dc3734b6cb105dcc5e5460a23 : SUCCESS in 53m 19s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1afe4d497a5245dab7fef421655eb1b9 : SUCCESS in 1h 41m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/baaec7bf63764c1783629936d8a2218d : SUCCESS in 54m 09s","accounts_in_message":[],"_revision_number":19},{"id":"949ddfebf9f621b5b425e72717730561b4da0158","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-09-03 11:13:11.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"de967a0c1269c8b43b63585ce6ebc52822971161","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-03 11:15:08.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":19},{"id":"ba6aaf50642b16e2159d5d618ab05160c28a6b06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 13:44:36.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac5d28931f31401181a62abeea1a325b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3cf8c90258e442f90e3f89cfdf17534 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba60be9a618340a7896c271c51f1fe84 : SUCCESS in 7m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dcb796867b9649c0b677fe2780421da1 : SUCCESS in 9m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae06e7e7b8354b198357a5438ae02844 : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db48e71856840f6ab2aa26a5dfd4060 : SUCCESS in 7m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/4bd789e9e1e84cc28d0d0c3ab37f523a : SUCCESS in 57m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c609d7a2540449a8be042c1aa28b2d3e : SUCCESS in 1h 00m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69a9b96c796e4aefbb7c26fc6723d285 : SUCCESS in 1h 50m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03f41f33433c4d83b7caaff0a2707453 : SUCCESS in 1h 06m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/faedfd0a3c79400ca18e4aaa6c4db823 : SUCCESS in 19m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/752b273e745245cab5a9415d2b265563 : SUCCESS in 3m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf29dbf51e6148dfa5ff999b74c6d9a6 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1e5f844ddc93401e85ced50305c5c850 : SUCCESS in 6m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bc2fe607f7c144a1bfdaed5ddd304439 : SUCCESS in 5m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/79af7be956544acdafb1f299d571e7c8 : FAILURE in 25m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/11d7ad3b5e974cbcbb03e9fbcad6b224 : SUCCESS in 3m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8ad49ee4f3d541dbb9c3d34baa206294 : SUCCESS in 1h 10m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a2db19f6d8b443a5bf6b8dc57194b692 : FAILURE in 2h 26m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ef91288075f24a60b24a1d46440c93f1 : SUCCESS in 1h 38m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0d687b2ee0fb4c0f9e97bbd454b1223f : FAILURE in 18m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c002726613134db3b7686e8a3d63ff7e : SUCCESS in 32m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7974c65ef0d04363922e04b8ec353efc : SUCCESS in 1h 15m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/328541972c604ebe9dcf95339587b2b1 : SUCCESS in 1h 51m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ecce71be2a224f2a9c5b99dc4892d16b : SUCCESS in 33m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e9380ad7b3654ddcbe0315d0d27c9ddb : SUCCESS in 47m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fc603ae9f1d047098858cecbf62ae754 : FAILURE in 19m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c52f351693494d37a43eb5e62c7260a8 : SUCCESS in 1h 30m 35s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0068975d5c44dd4ba65b77967eabd47 : SUCCESS in 1h 44m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b42fd9e70e74110bbd31fbb72606fd1 : SUCCESS in 41m 04s","accounts_in_message":[],"_revision_number":19},{"id":"d3c871a681c76ca123294a6fd145e823a569a0ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 14:08:07.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b366dd2a1bd44f368a9d8bad2af7ea5b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/14eb6a2d0e8c42938a790a6d06b12d34 : SUCCESS in 26m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/17de78355fc04c28986388bfe5c9fb25 : SUCCESS in 25m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/13b8569c7d244cd99ed811b10dcf9013 : SUCCESS in 25m 02s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"27cbdd66989256908145842b01ce6c689c0b927f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-03 14:36:56.000000000","message":"Patch Set 19:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1799/.","accounts_in_message":[],"_revision_number":19},{"id":"720021ab79ce0015f5909a6e442da0e28847a9b8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-03 15:01:32.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 1h 24m 42s","accounts_in_message":[],"_revision_number":19},{"id":"fe80a7bd2322ff002a947345f55a2414afa9bc59","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 15:08:05.000000000","message":"Patch Set 12:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/413/","accounts_in_message":[],"_revision_number":12},{"id":"ebe4bebb018082b918714bcc02d5e8af8ae98ff3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-03 15:17:08.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6b9308bfd43345fc85701484c5214b90 : SUCCESS in 44m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8cfc0ffe41da455d866ac767592f3b62 : FAILURE in 55m 52s","accounts_in_message":[],"_revision_number":19},{"id":"39a83063450a9ada7c570a3cf520ccc07ff79f50","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-03 15:42:08.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"2644351902f9ab140b2a33c0cef1970d3f308f0d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 16:04:32.000000000","message":"Patch Set 12:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/12/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"1551fd16883a4c1acd9e93db2967f1df11ebf605","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-03 16:43:23.000000000","message":"Patch Set 19:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1663/.","accounts_in_message":[],"_revision_number":19},{"id":"6f49caf87bbb5f0a74152901eafbe741f59870b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 22:40:30.000000000","message":"Patch Set 13:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/421/","accounts_in_message":[],"_revision_number":13},{"id":"d6f33d2c15f8837482455d2411adb6444a7c1ee3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-03 23:14:34.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"0a3febe457873511eff18301500f2ea4b0023578","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 23:39:08.000000000","message":"Patch Set 13:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/13/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"05befdab2657c01bd11afadc16f652c426bb1933","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 23:39:10.000000000","message":"Patch Set 14:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/422/","accounts_in_message":[],"_revision_number":14},{"id":"6a0f8419b69d1140c1b639768693e278ff5242cb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 00:38:29.000000000","message":"Patch Set 14:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/14/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/14/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":14},{"id":"7aeed21dc6c0c3b0ef2874807145aa5d6a381c16","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 00:38:31.000000000","message":"Patch Set 15:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/423/","accounts_in_message":[],"_revision_number":15},{"id":"742a0bfb51a8aa0ab16fbcf74f4d940511a6cf5e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-04 00:47:22.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"54e7045b07a60d23c5bf5ada3106af5f3f6937e5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-04 01:03:06.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"adeccb0aaac0f2c03d562431f52381a920e4a646","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 01:40:12.000000000","message":"Patch Set 15:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/15/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/15/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":15},{"id":"7567e43ae60be878a623b80132504c5ff77a89e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 01:40:14.000000000","message":"Patch Set 16:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/424/","accounts_in_message":[],"_revision_number":16},{"id":"e3f78ed9bddc1e01aa5adfb943aee7787ededca3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-04 02:33:02.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"77bec04539f01d94b5988e5a40c100defd8282cd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 02:37:52.000000000","message":"Patch Set 16:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/16/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/16/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":16},{"id":"b2168aecf9303987581ad246d6d5d8ed169db400","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 02:37:54.000000000","message":"Patch Set 17:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/425/","accounts_in_message":[],"_revision_number":17},{"id":"4556e440543bc4f7f66ed9189d007a1a88eb3599","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-04 02:58:01.000000000","message":"Patch Set 17:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"f6b27fdb48ae49521cf92b0cdf12e998b22b1e90","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 03:32:30.000000000","message":"Patch Set 17:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/17/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/17/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":17},{"id":"e1284c27c24e742c48d3000e8ad276721e9bd81b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 03:32:31.000000000","message":"Patch Set 18:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/426/","accounts_in_message":[],"_revision_number":18},{"id":"2f18add6bd14fa74d189e0bb47f5f750cb17759a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-04 04:21:42.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"200c0b68b1b3d9c1e0eb695b3b45f81b02a5086c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 04:27:05.000000000","message":"Patch Set 18:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/18/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/18/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":18},{"id":"e9d1b9247131aaab2ac498b5c5c5e41aa911a5ea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 04:27:07.000000000","message":"Patch Set 19:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/427/","accounts_in_message":[],"_revision_number":19},{"id":"cab2305b0ef721cf3bd7584ac19e6994d432f2da","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-04 04:48:09.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"8bf4cad2757ba94f2d17213e22a164111b1f99e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 05:21:32.000000000","message":"Patch Set 19:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/19/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/19/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":19},{"id":"eec4a7d3cdaf396dc00b0602b44e490e2fc4fdad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-08 10:47:16.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"935ed83993e02467bd702ae5df83f901c7d8d305","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-08 10:47:38.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"4c1f6af45e9b41c6dc5fc656c25937bf4ee710ef","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-08 10:49:08.000000000","message":"Patch Set 20:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":20},{"id":"f67677b304308004e9193776dc085f88c5e1b0f8","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-08 10:50:54.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"80917a15ae82ca1c941ad05d3e2fea93fdee8dc7","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-08 11:01:18.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"18c3b52d73e54e6670291030865cbc7447f90d16","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-08 11:04:15.000000000","message":"Patch Set 20:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":20},{"id":"596ca43dd35e2c27d4aba691ee45a67be0865917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 12:35:55.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dbd5b5aec66461a92e12475f7108f29\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bae2a503d3e4700b05bee13a97b31c6 : SUCCESS in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25402ba7fac84bd1ac85238918fdf671 : FAILURE in 10m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5de4c8664d9246228e86d4ed252d8489 : FAILURE in 10m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe1cd15b30c245a6989ebe6533daacd8 : FAILURE in 14m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14804f5748ce4890869923ba4e2bebd5 : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/aac28d06628243a490f765f184604d4d : SUCCESS in 42m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59a78aca1f544fbc871c588fcbd55868 : SUCCESS in 33m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6e299a21a254d87ae57317a67111519 : SUCCESS in 1h 22m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c10ef420a9d84fad946e363c8876b578 : SUCCESS in 1h 04m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3be09f9e6671494b83d8ddda1684f418 : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6df61e0a58c1462c91ed6dc07e33e31d : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/db51e74533b64dee8e53223e65012878 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a947f50a9f1f44829d20a75c866f04fc : SUCCESS in 4m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8f33786af72f42dbb099d6900a44861f : SUCCESS in 8m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fc5271a70b6142df8741ed89658ba9ef : FAILURE in 36m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c3c52ba4f0d4c80b3a3303965fbcae0 : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/707001542f6d48e7a88a1c7b2bfe86e0 : SUCCESS in 1h 43m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/135dce8ef98640ef8650af8b0feccd41 : FAILURE in 1h 38m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6141dc296e864d34aeea1dda30a22c3b : SUCCESS in 1h 43m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dfd755dfeb784fcc98170e5e4e13241a : FAILURE in 11m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c78cd1bbbd1349f888f9ede7184beab8 : SUCCESS in 34m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/50678c9acfa84ef3b8d9e65a3a8be187 : SUCCESS in 1h 08m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/152cfb0d6a5647ae8bc0675747dec51e : SUCCESS in 1h 23m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/def9cbf65878448f8d52940dbff01eb5 : SUCCESS in 51m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e8ec45b4ad34466abf603946cb8401dd : SUCCESS in 1h 36m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a680152cad4d4ec08bd81c2f2379f216 : FAILURE in 17m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/487fce22a86641a58ede38f32e31b87b : SUCCESS in 1h 22m 55s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a613d2eaaaa4351aef0dc6574aa41a7 : SUCCESS in 1h 32m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45b901dd2f0b4c42821f0ed6c1251ac0 : SUCCESS in 53m 50s","accounts_in_message":[],"_revision_number":20},{"id":"071114bde8bc51553549647445e4bfd1075b6879","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 12:36:21.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b915963d4344f00879525da05ae8def\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8bc459841454ad7807fe436d815941d : FAILURE in 10m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/589c3c31ed8b4e86a26e4afb4f95e93d : FAILURE in 26m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/efa6662100e9465bb73d2af362e47674 : FAILURE in 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"43f6775af696b32ab27a0748bd62d235368a632c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 12:58:04.000000000","message":"Patch Set 20:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/496/","accounts_in_message":[],"_revision_number":20},{"id":"4dec9caff05a7afd61a3b9e6797452f76831eabb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 13:17:56.000000000","message":"Patch Set 20:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/20/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/20/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":20},{"id":"887cf51c620c49c87ce467eb43c3b7a511df8bf6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 13:38:49.000000000","message":"Patch Set 20:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/498/","accounts_in_message":[],"_revision_number":20},{"id":"d75af1505a6ff0ea79472de6e99704d87b587edd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 13:58:37.000000000","message":"Patch Set 20:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/20/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/20/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":20},{"id":"1d12708518270fa79908d80b70eae315377a46f0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-08 16:23:34.000000000","message":"Patch Set 20:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1716/.","accounts_in_message":[],"_revision_number":20},{"id":"75b2c71b76c0f9df10dba264976761f842035546","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-08 16:37:13.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 40m 08s","accounts_in_message":[],"_revision_number":20},{"id":"47a6d6144e204829b9e7219700a71c1663afd9ad","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-08 16:47:18.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8083e12b36ea4522930b8825b6649682 : SUCCESS in 47m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e3b1060f1f78442295d8a37d9c9bb5ca : FAILURE in 56m 58s","accounts_in_message":[],"_revision_number":20},{"id":"ddf72f34f37d0e7fc6f1090a3e83bcfe93950b63","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-08 16:57:22.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/958465 : FAILURE in 41m 35s","accounts_in_message":[],"_revision_number":20},{"id":"821e549473a380e393a132a67447a971a061c415","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-08 18:18:30.000000000","message":"Patch Set 20:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1852/.","accounts_in_message":[],"_revision_number":20},{"id":"027cd55ed082830b473960e538e748fb17421727","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-08 19:17:15.000000000","message":"Patch Set 20:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":20},{"id":"a1bf135fd45b4f9ad87d95ec2978f41a3fe7633c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-10 15:20:15.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"551fe2827318f2b8ce550453c678df442edcc8c1","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-10 15:20:58.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"05c62fcb5e009c5481ccf86d7f56d81ca4b2ff31","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 15:42:29.000000000","message":"Patch Set 21:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/542/","accounts_in_message":[],"_revision_number":21},{"id":"97361d0e8232caee99af1d95cae293608e9e2d86","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 16:01:15.000000000","message":"Patch Set 21:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/21/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/21/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":21},{"id":"d25b1be9821ed9077cbba93920690886668a7a81","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-10 23:50:12.000000000","message":"Patch Set 21:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":21},{"id":"8aa4c15fa73d95277dc63cbb37ac0320b9a41de5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-11 00:02:47.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/958465 : FAILURE in 40m 25s","accounts_in_message":[],"_revision_number":21},{"id":"1c6a536c74cd68ef3d0ca05c1cfd3722395fc18f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-11 01:43:52.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 39m 01s","accounts_in_message":[],"_revision_number":21},{"id":"b70eaba8723f7910dbae8d4fb6e99107808ca194","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-11 02:06:20.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/43a5e1b8814043e39c9ea475f1aa1093 : SUCCESS in 48m 18s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a18df97cdf7b495bbe2747495bb82901 : FAILURE in 1h 00m 43s","accounts_in_message":[],"_revision_number":21},{"id":"7082ae45a70f91bfd894154ef488be59a5090cf3","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-09-11 07:24:47.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"488ca92a4c1ee314873803ddd1256014e762772c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-11 10:27:27.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/76d4d8718286499bb04426d50f986a2c : FAILURE in 46m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9123480e0a4c4652a69986739460baab : FAILURE in 1h 06m 09s","accounts_in_message":[],"_revision_number":21},{"id":"82ffecc3932f2f00fb694bfa72e765357c13946e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-10-31 15:32:34.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"089674c5639b406bf5f6f4635269e338e2c890c9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-31 15:39:12.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":22},{"id":"dbe63130249a1e6dc413b6364f8f5a728cdbc1e9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-31 16:31:22.000000000","message":"Patch Set 22:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":22},{"id":"81a9a1d5e69efcf721849ef7d23c1e1a5be0ec43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:59:57.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee73bfec9fec43c78eeef138517d925a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b348e8d1d1246da8c2647d16127fcc0 : SUCCESS in 20m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c81a4b4d4e9241dfab745539a70343fc : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/afc741723f6f49bfa00ef3f1053088b8 : SUCCESS in 27m 57s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"263be90853cd2d51b3b45703603ccf70df0ccd0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 18:54:45.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3c5ba2194324d379b107f2d902d473d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d1c33d09d84c659cc720b8f4753a4c : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65b00dfaafe349928da651793ac65b97 : SUCCESS in 11m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b87852f3fac94e0ba61550ec0512e31e : SUCCESS in 14m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9ccc5927384394bf0b48030744c3de : SUCCESS in 7m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/7ffda1ddab924ff1bae89bc92f34d311 : SUCCESS in 1h 10m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df62d28334364e45a4382c30ec019e14 : SUCCESS in 53m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ebd88ead6e68447481305c9d0b73cbb7 : SUCCESS in 1h 52m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb0bc87223c44574a3458c01265d17b5 : SUCCESS in 1h 01m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a4f5178fa07244cdbf6ee7391865faf5 : SUCCESS in 30m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/21fe2cb13ae1406297863a71fcd4aaa9 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8ce8893e308f41aab05cd611baf59ea8 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f6619aa41779476f90e4af2901811e18 : SUCCESS in 6m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/24feb10aa6b84dc792465f35eee156f6 : SUCCESS in 9m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4688563a3cbf43448a3b0595f4c35de5 : FAILURE in 19m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b9aed0d1a594595b53eaa8c3dbb7232 : SUCCESS in 3m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ed84b83287f7443c8c6d85d5878cba3e : SUCCESS in 1h 45m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bc9ca75f86344976be26d02592c02035 : FAILURE in 2h 43m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/411980af1ce14b72b979bf31ba9c0c27 : SUCCESS in 2h 04m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/024d850d9a0344d2bca24fbfe3d74aa2 : SUCCESS in 1h 59m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/929cbb8198d24321ab6f39bae0dd2905 : SUCCESS in 37m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dce50db8cca6410e81c30adecbf57fd5 : SUCCESS in 44m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/69433c0afb704c068063f311addbd978 : SUCCESS in 1h 34m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d6f5ad64784641e9bc8db47d5eaeaef4 : SUCCESS in 54m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/71966ce2ad614b2d84469faf19044614 : SUCCESS in 2h 11m 54s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/1f391f8478dc49c2af965bda86603a67 : POST_FAILURE in 2h 13m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/819bd7869c5244e1992ef861ebca705f : SUCCESS in 53m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39c85a7a0f3749c88cbad61f6f999419 : SUCCESS in 1h 08m 45s","accounts_in_message":[],"_revision_number":22},{"id":"59899169b4521f890eb8aaea386d1cefb5a13706","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-31 19:11:31.000000000","message":"Patch Set 22:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_10_2025_19_04_48\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":22},{"id":"b2039df5b61ed0318181f765057fd051e379aaf7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-31 22:39:15.000000000","message":"Patch Set 22:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2252/.","accounts_in_message":[],"_revision_number":22},{"id":"ac9325c35cf6a9401dd301cf4d920b1905ed7df9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-31 22:46:00.000000000","message":"Patch Set 22:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2366/.","accounts_in_message":[],"_revision_number":22},{"id":"28e15c8137436d1cfb4f18be506a354f7570c70f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-01 01:18:57.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 36m 04s","accounts_in_message":[],"_revision_number":22},{"id":"0e90dffa1b4e45f1a1d86224b2586135841a35ef","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-01 09:26:06.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2344f8a951d74d8d9035af4e6aa846b2 : SUCCESS in 43m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d1ad8858aa014b53ae8c1acff9b78e6b : FAILURE in 1h 08m 35s","accounts_in_message":[],"_revision_number":22},{"id":"2c3d63a6bc723f327ecb1fc32d7eebce6c5ebb83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-05 13:47:46.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"c9da53d8c64d05b499db9efe810aac5c427679f4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-05 13:49:25.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 938771,74","accounts_in_message":[],"_revision_number":23},{"id":"19a9f526488717d26b2ec76da9d80cb0f85ae64e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-05 13:51:29.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":23},{"id":"f4fb612ae7dfcab3cc1faf77cf45c5ddd55fa491","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-05 14:50:02.000000000","message":"Patch Set 23:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":23},{"id":"26e70183da641ae6a018248784737f393882d8db","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-05 18:19:13.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 34m 50s","accounts_in_message":[],"_revision_number":23},{"id":"97b0f9f27a2158ccaa7ee0f576cda308640fcf4e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-06 04:06:41.000000000","message":"Patch Set 23:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2429/.","accounts_in_message":[],"_revision_number":23},{"id":"1dfaaedb59c572191444d0ddfeb464a1641ff5b9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-06 04:06:45.000000000","message":"Patch Set 23:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2315/.","accounts_in_message":[],"_revision_number":23},{"id":"d5dcfb83523090c6478e8e1f5896146c0ab95d29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-07 11:46:00.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"99e9f166014f1b4417fe80a6dc3e578e1803d51c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-07 11:47:17.000000000","message":"Patch Set 24:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 938771,74","accounts_in_message":[],"_revision_number":24},{"id":"f56e0eaf95c9a658fdc4f018c59c168f111d3954","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-07 11:48:55.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":24},{"id":"33afa85a81e76f63bb2fc74ed5bc3052bdcacdcf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-07 12:46:51.000000000","message":"Patch Set 24:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":24},{"id":"ff4743c24af5328321ed87daef09392460b68feb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-07 13:04:40.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 34m 52s","accounts_in_message":[],"_revision_number":24},{"id":"f68ea5adde4f6d26e3e42b00d6335d3862ae282e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-07 13:47:00.000000000","message":"Patch Set 24:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2339/.","accounts_in_message":[],"_revision_number":24},{"id":"49b6e790301b49373561328c6f4b1d1d041d1ec9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-07 13:53:25.000000000","message":"Patch Set 24:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2453/.","accounts_in_message":[],"_revision_number":24},{"id":"30b41d9b4411403ab01ecf272ca57e2e78137d81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-11-11 10:20:43.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"6fa3478d9e2d511dd5dbaa597ae9e7678fda108e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-11 10:26:27.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":25},{"id":"e53108ccab9ae7c5120d34c636970c6a4a59cc4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 11:19:39.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9672320793b48db996bbb3bdbcf3db1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ae588bd4cab42319c0d6a0f02a026c7 : SUCCESS in 21m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4019705d2f4a460a9c7c7056b30f1b7c : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/049bd73e36894f64841c06b800bd630e : SUCCESS in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"46050a0298817045d896518bca00562be983dad3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-11 11:55:59.000000000","message":"Patch Set 25:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":25},{"id":"5b642e58b0cddd0e08c7754a3ab2354211f55aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 12:39:07.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0acc9047d324d9bba65dc779dd0585c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4da50edfff04701870e32390a0c6bbe : SUCCESS in 7m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74b6894c2abf4019a612db8382425e8f : SUCCESS in 12m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88fd0507d1614781a0e5b8063c7f8765 : SUCCESS in 14m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09f6ff319474a4eb41ae3be119065d9 : SUCCESS in 15m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/098bf419e38648a0a4f6fafcb40038c2 : SUCCESS in 1h 01m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/353b119f15c643cab8a4cb2a36461dc1 : SUCCESS in 1h 03m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5970c50f753d407d9211e9f0c0a823cd : SUCCESS in 1h 41m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/583c78ae16ed465d9bb15954d111ff24 : SUCCESS in 58m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2af74fc219ef4b3a99d2f9bd5d6230ae : SUCCESS in 25m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fe9c2231ba834702b1589ef5a48224a9 : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1eaa7f506ef48b2bb2b9af44cc24db0 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e21b3e4a69b34625a2775e3767d9130f : SUCCESS in 5m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9b9ba72d0e2f4900b79c45cb18e492e1 : SUCCESS in 8m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bd0c96d553954c09adeaea48ed767599 : FAILURE in 38m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb4fb32c1ed64f71974a95969b220529 : SUCCESS in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/678d710111fa470dbf80935cb71ffda6 : SUCCESS in 1h 42m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7fbb836fec724806924e11260e75501c : FAILURE in 1h 47m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bc9d5fe2574d41ca99153e9669ef2073 : SUCCESS in 1h 49m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/622eb2885656432389486b1d0c8578fa : SUCCESS in 1h 53m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/aaac648b5c0042afb6c210d08c770c10 : SUCCESS in 33m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a2b91c352a764b8fb50e8a69da7e9578 : SUCCESS in 57m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5ecafbc930c24accbce8b2cdb75cbef2 : SUCCESS in 1h 46m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/34dca42d88c34e1ab7542e2bf9ee04e7 : SUCCESS in 43m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0cb7f816c5774999862efb563bfc7a7c : SUCCESS in 2h 09m 43s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/8b8e1ad6a812462692ec37e13a839fd8 : SUCCESS in 2h 04m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ca0d7f2d689b4e988f4400091e4bae90 : SUCCESS in 2h 00m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0949b645a25d4d1a822761fb01115bf0 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":25},{"id":"c7817929c05d89a7608a599dcb962a89e1443732","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 14:20:02.000000000","message":"Patch Set 25:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2508/.","accounts_in_message":[],"_revision_number":25},{"id":"269023ad7e6c32112c395111191a89f285ba617e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 14:46:54.000000000","message":"Patch Set 25:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2394/.","accounts_in_message":[],"_revision_number":25},{"id":"397f1d457dbf009aa80c4bc0953f600293cdfa11","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-11 19:33:33.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/850ecabe554d43e8b121d2e625de3e3e : FAILURE in 48m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/785fc149e2c44618b02a6f9a41d1fd01 : FAILURE in 1h 08m 00s","accounts_in_message":[],"_revision_number":25},{"id":"de5a85ef685c913adab33b2db4c4dc992679d288","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-11 22:07:13.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 35m 56s","accounts_in_message":[],"_revision_number":25},{"id":"fa7d9b46d350d222a2207fc788ce8d1327f60b6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-12 09:35:16.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"6839331c4e6122739f3431999944dba4e8f6c9b1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 09:37:37.000000000","message":"Patch Set 26:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2530/.","accounts_in_message":[],"_revision_number":26},{"id":"6f81bf6e46c9d779e3162f0e94c4fb5f90cde748","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 09:37:57.000000000","message":"Patch Set 26:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2416/.","accounts_in_message":[],"_revision_number":26},{"id":"c05642e916013f17809190a362fdd018885a2f11","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-12 09:40:52.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":26},{"id":"eb342f589cbefc61e3936f3cb543ed26cfc47067","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-12 10:17:46.000000000","message":"Patch Set 26:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":26},{"id":"4e30d7aa2f1cfffba635be9b7909c8ace3b90ac0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 10:21:48.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7bdce2c05ea433bab99c8a4932196ce\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74c31495bd274374847b0f3462e2e0b9 : SUCCESS in 13m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4682d2238aea48f3beeb39263c26baaf : SUCCESS in 14m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c6c1484a5b5a43f5802c4daa732eed7b : SUCCESS in 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"04ad169a86f21daca197665e63468848230b20a3","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-12 10:23:35.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"e9ec74fc2233e5c2ea03d9d8e7d1442f58eb2e9e","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-12 10:23:44.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_13915\u003e.","accounts_in_message":[{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}],"_revision_number":26},{"id":"77d447603c86b59b30d5733726eae570b51964f5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-12 11:20:37.000000000","message":"Patch Set 26:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":26},{"id":"a6ac00a6001efbd7092a00560d4e9ebb4e363708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 12:41:36.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfa7e4c044844ee5bd6c417f5ad5e3ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08cecc2a6a6b46df8369429b5e1bdeda : SUCCESS in 8m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f911455432d4b7593d3b2e339047363 : SUCCESS in 12m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7ef6f85219a4c51a6a0c0278fde8700 : SUCCESS in 15m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa7b32954d0475c88b412d854ce6ae8 : SUCCESS in 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/4274741de0564b4088de7bcce549a455 : SUCCESS in 1h 01m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7417af19d5234c338273db2fe67efe1b : SUCCESS in 59m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47dbd4677c68459594ca4901bab768fa : SUCCESS in 1h 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/560b59e3aeeb465eab63206916d5311d : SUCCESS in 58m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8fd125d2560b425f9322f30c91d8b11b : TIMED_OUT in 41m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/821bb09923634f328b6323cb05b4d526 : SUCCESS in 5m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf4e06ad55644ab7ba9f0eae009786fd : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5edcab666f814417aa6a51e0b5128751 : SUCCESS in 6m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a9821eab1ca14f3aa83d32a7c29c42df : SUCCESS in 8m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a03d58ced9d44cf8a86eac077ce991c9 : FAILURE in 42m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d017df80b53426395c043a769439aa1 : SUCCESS in 7m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b078bd39114a4508a806363f2a95f2dd : SUCCESS in 1h 58m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5ca0bb886dc540c09c40c281f264af77 : FAILURE in 3h 00m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2a22ef9e26544707a1196530c8394a14 : SUCCESS in 2h 20m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2b871ac8883c438a887a36ea8b52022d : SUCCESS in 2h 45m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cca645c290d545bc8e5b7969fefa6efc : SUCCESS in 26m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/af62c6586ab443e493c6b1c11b1a6a6f : SUCCESS in 1h 12m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/189a5dc8b31b4253b53ca636099d057e : SUCCESS in 1h 30m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f2c14786784f444dbba678062eefe0b0 : SUCCESS in 47m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/19d2e980f23749aaa520214f7bcd3168 : SUCCESS in 1h 57m 38s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/5b32c8a3042f404d84a575f689b276c3 : SUCCESS in 1h 57m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2d6e1e1193714112bab98618489818db : SUCCESS in 1h 39m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36e89229e4f1419f9bb62dcd60073584 : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":26},{"id":"06e8a9017a0713646d94af0c6ff317ba0fb1cdc5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-12 17:35:52.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 37m 00s","accounts_in_message":[],"_revision_number":26},{"id":"56ac6b38705b6c8827a17b675028b7f9d9483ca7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-12 21:09:12.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bf9c34c2d99e42789ecf300b7c124ae7 : SUCCESS in 46m 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/00ee998c1180497f9bda1e4692f192b5 : FAILURE in 55m 39s","accounts_in_message":[],"_revision_number":26},{"id":"213647e4b8f3e8af612664fc11bda44b0b235a75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2025-11-13 10:22:02.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"233a0e00c17b16e0a724ce58ae726132e3575d9b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 10:28:08.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":27},{"id":"243211f07ec592bbcda0c36a46e9937d92b59ea4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-13 11:23:57.000000000","message":"Patch Set 27:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":27},{"id":"c37b37db2b8f3848bb054d06e5989e55e952419a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 11:31:38.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ac53b5950df49bfa8ee41293e1dfd59\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c9ead2915f654387910d9465df7fb1dd : SUCCESS in 20m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/111e691551e84ef0a6bcf0ac4712c097 : SUCCESS in 14m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aeef01c51d2c4becaa7023470d3491c6 : SUCCESS in 25m 10s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"2b5903fda8036e89c271d04d6027a07cc1548707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-11-13 11:52:09.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"19ab1862db526c3beaa0e618f7611a6872801d25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-11-13 11:57:28.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"49108308ae5865d4f2eea7aa280f20b6d6e47f07","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 12:01:50.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":28},{"id":"3f50bfbb72628dbb003a7e3c6709612fb47ea9b7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 12:18:57.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":29},{"id":"424f0e38c91b7ee281368af31a78f8649a6aff99","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-13 12:53:32.000000000","message":"Patch Set 28:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":28},{"id":"decf03d4ae995e3b1ff0fafd2cdcac9e04a19415","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 13:49:28.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57d6d4e294384a5689baa2ba77618ee5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/06334f1fecb44636b33d0901cac720d7 : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6d7f090be1d34b079c0f0962a8e6f15d : SUCCESS in 23m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/92cdfba38a7c410a9e830d6f8de20eb5 : SUCCESS in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"dbc7bea7732c5062584ff09a74a2d5b675f4bdcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:15:41.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f12589725ddb4b8a8104b3b099aa1457\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f59507c9db1e4e079968ebbbbc0dafbc : SUCCESS in 8m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d5421c4ced7442497483c436bd43006 : SUCCESS in 11m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61fe70f8350a407bac3202103ceaa30f : SUCCESS in 14m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5fce153baba47cc94bd80e997e4222a : SUCCESS in 14m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/fb6994bcd3214754b94c5cf2e7014304 : SUCCESS in 53m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29adfe4cda2b46018b12572f2af2db99 : SUCCESS in 1h 02m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b86bb4f620e4e78873413b2d1ecf587 : SUCCESS in 1h 24m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec7eaf3b80b1413aa32496a43225deac : SUCCESS in 53m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ad6ec330aa5344238236bac882f6f36d : SUCCESS in 23m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/87a7cd676e12418fb30e78d8b489a40a : SUCCESS in 5m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/545ca2e33f104278803d93f328e231c3 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/78c28a4a0102442887651675b0786e1d : SUCCESS in 6m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d2ef720fdee246599496d4ebd29ce5a9 : SUCCESS in 7m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/981ec2e12f0047c4a52b937c61f949cb : FAILURE in 40m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f2f8cbdf2984d6cadbabd0c25518b45 : SUCCESS in 7m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0282dc49855d4efcbbecde32c4dcf9e2 : FAILURE in 1h 46m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f5f8e9d420c6447b933c956ef756d840 : TIMED_OUT in 3h 05m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f55298443550439e9cc20d2667112783 : SUCCESS in 1h 54m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/462c8d2df5024efe8703523dbb2ad12a : SUCCESS in 1h 54m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d5c13acf655f4040b7c57730c31f99d7 : SUCCESS in 35m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/07ad78f1102948938c6a47bd4c8f58b0 : SUCCESS in 1h 03m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/73843a226aa343968275a91fe6ee75cf : SUCCESS in 1h 32m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ee3e6576bc724f82996fb03063001a94 : SUCCESS in 52m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fcdd5607b0354dc4bf23093bfca25771 : SUCCESS in 2h 10m 27s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/b7747e6b4fbe4290a899a242d920e163 : SUCCESS in 2h 32m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a54859c31f04145a0179b5db2f33f6a : SUCCESS in 1h 41m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5bd556058db4a88afbee1f2b47c2694 : SUCCESS in 57m 34s","accounts_in_message":[],"_revision_number":29},{"id":"0f4e2491f966c3b8b2f33fcc6757a57e26d9506d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-13 15:30:24.000000000","message":"Patch Set 29:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":29},{"id":"05f08b221719b7ef4067a446db6e3bba94b26f69","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-13 18:05:37.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e30147eed3274c33a76d4f99e1266894 : FAILURE in 53m 20s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/21ba7925e2664ba2ad550734438499d7 : SUCCESS in 55m 32s","accounts_in_message":[],"_revision_number":29},{"id":"77764e84d85b080276b048618b65fa09d83aeba0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-13 18:32:16.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 35m 00s","accounts_in_message":[],"_revision_number":29},{"id":"1d988dc1b86dfef3bb8a18c890eb2df186c17f20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2025-11-20 05:54:45.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"23ce2c5a94e373b159bef63eead03ee99edb99e0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 05:56:20.000000000","message":"Patch Set 30:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2529/.","accounts_in_message":[],"_revision_number":30},{"id":"5aa12161257ee53bdc9f144785366ef8440c6b75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 06:38:10.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2a1d7bb9f354af8979ac397e6c966e4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7fcb9b50c3e446f7bbf69d2a93101661 : SUCCESS in 24m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8620da588fea4b2897197b0d44f0be80 : SUCCESS in 13m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b00ed196cbd241fa84046bab220d4ea5 : SUCCESS in 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"1df46f807f6c1fb7b0efb8be181b505bc024ee89","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-20 07:11:44.000000000","message":"Patch Set 30:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":30},{"id":"7037de3118f69f3cb21f5da92887692512ecc88d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-20 08:09:44.000000000","message":"Patch Set 30:\n\nBuild Successful \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/28/ : SUCCESS","accounts_in_message":[],"_revision_number":30},{"id":"ffe1b2ef572cacdeead196ed589b2669d094d90b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 09:12:12.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a02959036e034886b69703b94cd27cce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3654acfe0d6945c5bbf85c908d12eb28 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/408debdb0d014c0e80e3ffa6b8649008 : SUCCESS in 11m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/089c850ad6c140daa992c64a9b54ab3e : SUCCESS in 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe2b5136d6c436cba9e57064dd9cb1d : SUCCESS in 14m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/3bf06e11da204d829340da804644820f : SUCCESS in 41m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5e87c89bf15c43fc9fbf5e779e4117d0 : SUCCESS in 43m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/602ee0124e3c457ebe51b731f0524add : SUCCESS in 1h 32m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0b1e5bdc583476e8773004f28593c7a : SUCCESS in 1h 01m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4d7fffee805f4ae4be7755a8cbe23c49 : SUCCESS in 20m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/96bdc9bf482b456d80c9489afe1cbe60 : SUCCESS in 4m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/345d0f33bc38451bbca864b40f77233e : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/87cfe6bb9dbd493d83f8411cc864b775 : SUCCESS in 6m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/97b7a79224594956a80069d2f8263783 : SUCCESS in 10m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/59a380a2bfc743b5b9c6f50f08d81cd2 : FAILURE in 40m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/261f692d14ff40c6a1e585bcae4a7938 : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fb57d5a785194463ad25bdb9afc8d625 : SUCCESS in 1h 11m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e87df80c657f4a3598489717f4ca2ab2 : FAILURE in 3h 02m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7cf8ecb2d57b431f983014ee5ab3828c : SUCCESS in 1h 51m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/55d625b7b2a541c0826bdc373212b037 : SUCCESS in 1h 31m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eaac405d018b409ca5f1439855c03f0d : SUCCESS in 27m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2cdf9d17220d46f3b4f9303af73282b1 : SUCCESS in 40m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ddfbbb29b2e9431db6b7549a3c84dc50 : FAILURE in 1h 37m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c3d555d387824886887f32d80e9174ed : SUCCESS in 29m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4bc52a372ddc40d5910f70e0259b46f1 : SUCCESS in 1h 11m 57s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0c50f540808b453aa0eea2f9971fa6a9 : FAILURE in 1h 25m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/016f23505a424d7784ff24fb534d928b : SUCCESS in 1h 06m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/337ca6195db8493caf228b1e62725192 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":30},{"id":"833d4fa8d52d620fe75a890d50051b9fc018546e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-20 11:54:54.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/09f177f6c6cb4161a172ab555ee6e5ba : SUCCESS in 47m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/75b01be6c5b34238b1685f30a5c7a211 : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":30},{"id":"868fa1d53149abc63eaa9e8c71c0ec9ca29d2c38","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-20 12:22:30.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 36m 25s","accounts_in_message":[],"_revision_number":30},{"id":"d949a4e1cbfb4f023d3de1979ba78fcf37dbac6a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 15:34:20.000000000","message":"Patch Set 30:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2673/.","accounts_in_message":[],"_revision_number":30},{"id":"158dff316e5c0e3cedaee9b13ad5c88f210d98b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-01-05 08:30:10.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"286ee7450889c6b58fa8f531fd437061382e8211","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-05 08:48:31.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7093151aadf04ad4906f301154213d98 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0a2f99ad82ec451581a8dec712a5fdff : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":31},{"id":"729268b55b27e93cf0ac74c1bc122566c8632b71","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-05 08:53:27.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1328d832bf234680a89f5d956e47e28f : NODE_FAILURE Node request 200-0000942312 failed in 0s","accounts_in_message":[],"_revision_number":31},{"id":"cddd62045b509f125c4a13457e9cc85c41985563","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 09:33:31.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb7a993ecd4b4a37af40166deebf6cb8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5ce61acce5af4382b39c4593da901510 : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e3ffad5440744a9b467e7516931ffb5 : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/58520c99103a4acc962e0e395805f5f6 : SUCCESS in 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"facd045163d7405f752d172990c2db121c75e8c2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-05 09:35:05.000000000","message":"Patch Set 31:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":31},{"id":"e6d33a44b2e5ebb2eae4ae5aa4f9ab5b639e512e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 11:43:47.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90fc3761f3d84a6bbfd2e1fb3eede818\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73e790d49e204904b5e224e29b4bffe7 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b6bcf62eaab4357b720396ed3908e8e : SUCCESS in 10m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1c8606ab64944fab606dcde925467ab : SUCCESS in 12m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278a01ec4df440dd964071fd4c895d5c : SUCCESS in 14m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/2f613fbc9fb74eeca5631f62a4fb72f9 : SUCCESS in 57m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/27f26ca70264469da64414b0c9ab5200 : SUCCESS in 1h 01m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/348b1f0109ba4717924b85ff8ef342ae : SUCCESS in 1h 31m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61f4c42b6cc44965b302a76e375129cd : SUCCESS in 52m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/33ebb990191142f3973efa3a282e418d : FAILURE in 26m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7686cb792268476ab021984eb2266d1c : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6093aee3bad846229441c6cb4d591bca : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e244563bbcad4b82b4a6c72aac41233a : SUCCESS in 5m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8333f960be554e69b103e5569daeaa61 : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/04efb4f486e547ac950b54f71616ca7b : FAILURE in 39m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/78778550bba1434fa07d1112e14a0d4b : SUCCESS in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/57241faa7d99474f8ea915b220aaabf2 : SUCCESS in 1h 38m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/45dbd4be816746d5ae82b78858973dfe : FAILURE in 2h 31m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5052547074c04cf6ab706c5badcdba02 : SUCCESS in 1h 38m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a9a0f3f1a31542d88c3a7090577ddfb3 : SUCCESS in 1h 55m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dd583a22493d4b6f81a4f527ca61d313 : SUCCESS in 32m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2db243c39bb4469cbe08e8dee95a9004 : SUCCESS in 1h 03m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/02d56a1ac5864599ad3a0ac6307a92ac : SUCCESS in 1h 31m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6bce3083b0d340aabad49d620a3585c9 : SUCCESS in 37m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/30e3f7bc546f4c4698dce99edf77eef9 : SUCCESS in 1h 53m 37s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/15a19b4208c14cbb8f38ec6f0f21a3f6 : TIMED_OUT in 3h 06m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da95c1df4132434aa148aebd77498bc2 : SUCCESS in 1h 43m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/709c3f5c3ffd4719a42ed5e10c556f97 : SUCCESS in 49m 40s","accounts_in_message":[],"_revision_number":31},{"id":"485524d11fdb20d09e0371083a5aa40b0d1cf723","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-05 11:52:59.000000000","message":"Patch Set 31:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_01_2026_11_45_48\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":31},{"id":"5eb57f590ed72c0c879e2d357718f771cfd3d553","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-05 17:34:21.000000000","message":"Patch Set 31:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_64/index.html : FAILURE in 543m","accounts_in_message":[],"_revision_number":31},{"id":"4215923f389ea0c9a0f6cf485ccca24d21c12005","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-05 23:39:28.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 36m 59s","accounts_in_message":[],"_revision_number":31},{"id":"4c8e0e0b3b841ebf3c8947efaf4c849e1471ceae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-01-06 10:23:31.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"a0871c82b21477251451e6d15747c49bb63aa515","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-06 10:45:50.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8b7b4b2001064397851defa6275d4564 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3bf8b9919f314cea866411b40113e82d : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":32},{"id":"f10094ec0e9b5c7211126d72ed995d0e68106ccf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-06 10:47:49.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/8dfbfc39885e4a29ae312aabb9b1a48c : NODE_FAILURE Node request 200-0000942335 failed in 0s","accounts_in_message":[],"_revision_number":32},{"id":"08cfc32788b0b0cf2d045eb04b5d8ce87b1b3a79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 11:16:29.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41fddb4ba63641c7aa227c5766585a21\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45f3c636f795435890c07b78ddaf7daa : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/783f22a33ed24800bab3f3e2617dfab4 : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dab373a68e6943eeb894b3e13de9a923 : SUCCESS in 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"a3c80facdca217ca56e34848ab19f7bfb7dac5ad","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-06 11:29:42.000000000","message":"Patch Set 32:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_89/index.html : FAILURE in 65m","accounts_in_message":[],"_revision_number":32},{"id":"9353a95d19f4fd183da53f3290a83ad4248c2013","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-06 11:31:25.000000000","message":"Patch Set 32:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":32},{"id":"610fb663600c74223655dd887adf4691aeb70bf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 13:03:09.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ac87b00fd3a44a09984e9e388a5ca7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3881bb8e10b4412ad8960c879185168 : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3fee1e2d5d54879a1fcbf3b81eed605 : SUCCESS in 10m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e945675e17d49b0afeace90ffb846d8 : TIMED_OUT in 40m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c592fcddfd04e75ba889b415bd3276f : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/aeecf4d841ac4baf990b4a42cf1b6318 : SUCCESS in 39m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/55a1f0dc868a455fb1aadb58b58e51e9 : SUCCESS in 53m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d2a975a920848af933ff72c29f83afc : SUCCESS in 1h 45m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac4c457452da419a9842451ce5bbf80e : SUCCESS in 40m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5079de4328ad4249ba770f8e7de7f219 : TIMED_OUT in 41m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4fc7342e1be54d28a3899dc3b9459e2c : SUCCESS in 3m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c73f1c568155407daf47fb3f084e9617 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3e7f10a3a8cd48b696dfaa0038563569 : SUCCESS in 7m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d6544e4257b9405db9961cceebab1ab3 : SUCCESS in 10m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/de930d7913c641c398405ce632e49063 : FAILURE in 23m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f836895018ca46ef803240a0e31ab7e1 : SUCCESS in 3m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b92395bbc2f415aadec737a01f8b0f2 : SUCCESS in 1h 09m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b76db3f035ee4c5195cf1ca4ba3301c5 : FAILURE in 2h 09m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1dce55bd14fe4e308f33dd2d4aaf67d2 : SUCCESS in 1h 45m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9e5ee735c1fa4aa8a60c1cf206c6065a : FAILURE in 1h 58m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/54564f97ab664ba3a23cbdaa4c4c86ec : SUCCESS in 30m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/27e19a483c144484bb13f50b127a6555 : SUCCESS in 59m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fd5b2143086849809bb543097041e459 : SUCCESS in 1h 25m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4d1e67a2bdc747a69608bf727e16c860 : SUCCESS in 23m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5212e61a733c427caf96af820c2bbbfb : SUCCESS in 1h 55m 00s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/4d92ec2bd15e4e2ea5ad5c1c1e85221c : SUCCESS in 2h 13m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b19d94d7c684722add1d4401cd7c8a6 : SUCCESS in 1h 14m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8713acf13b4743ae8644e434433e1159 : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":32},{"id":"52c5beb1b22ad8e1899fe5ac84b1fc05eaab2105","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-06 19:02:40.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 36m 03s","accounts_in_message":[],"_revision_number":32},{"id":"104f3a5eff43889b5290700ab064abdf414b9005","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-07 14:37:34.000000000","message":"Patch Set 33: Published edit on patch set 32.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"c6f30524e7a5de3295bf3b17ad8d28538014dddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-07 14:37:59.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"1abe245f210ae123e7cc5d4b35b4cd3821c3d9b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-07 14:39:16.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"e38ab7e482e61bf37690e9cb31a0c00c807a4d39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-07 14:39:40.000000000","message":"Patch Set 36: Published edit on patch set 35.","accounts_in_message":[],"_revision_number":36},{"id":"5624abfcc88462aeaa651ef32e50ec9f49d51857","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 15:12:58.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c0dcfcd08304be59ea92d0b8a7b7c60\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f3277d51d82a4e008af30a37f49fe955 : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e794f572e9a4456ab6fe7f8f252df5f : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/37743de8100447f899b7651569959eaf : SUCCESS in 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"fac032a7bba92c987041507c3d718a3999e3797e","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-07 15:13:21.000000000","message":"Patch Set 36:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/949006, https://review.opendev.org/951157 are needed.","accounts_in_message":[],"_revision_number":36},{"id":"137f05df03d44cd94102d49f36eb6718d3f922eb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-07 15:34:13.000000000","message":"Patch Set 33:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":33},{"id":"a3f6b347b36f000cdbadeff2c0e810bbca8bcf21","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-07 15:34:33.000000000","message":"Patch Set 35:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":35},{"id":"cae2d29ab58571ee753d95cfca21ecd301f7b269","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-07 15:35:35.000000000","message":"Patch Set 34:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":34},{"id":"9bf9297e954eaa76e5547b51075979deddf10e6c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-07 15:42:11.000000000","message":"Patch Set 36:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":36},{"id":"c96833e0a448e07a8cb35855ad93144b2d0b07c0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-07 15:53:39.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f8e4b64611ff42db83d792d5407f88d0 : FAILURE in 1h 11m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4393ed21c2744266aa4dd54143b84bf5 : RETRY_LIMIT in 17m 26s","accounts_in_message":[],"_revision_number":36},{"id":"f66449c4840b8c8296006ad718b708e21b0aaffe","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-07 16:21:44.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 1h 20m 16s","accounts_in_message":[],"_revision_number":36},{"id":"9c52d435d495c7cf2f0c3ae5082332df544f88de","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-07 16:55:14.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"4600f6781c9015b1a6730bdfa6686b44545e3726","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 17:31:11.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7635f92a28e42b8be8eb0c1703b133d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5533a674342f40ea9629b52f11da5f3a : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/985ce5742d5f44c58bdd5ea88f898fb2 : SUCCESS in 14m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dd6ed49fcc3240d581c1dc7164073aa9 : SUCCESS in 16m 05s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"6b8b7dddf1bb26e7d8f9efd1aded72c0c342014d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 17:56:34.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7242d94bbcd041abb06325b74ca11fcf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80004dfb50394dd890fc0dfd8c356743 : FAILURE in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4deb95f966c48ea9b15a8d90e4f094e : SUCCESS in 8m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dde785979e6f44529b8b5422840829c8 : SUCCESS in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb988c983414a58a7b93c22f10e883c : SUCCESS in 14m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/3ca1651a94e74c418f16e1ba5c49f7fb : SUCCESS in 50m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/72de66c480c2484b91a6475c3ffe1b79 : FAILURE in 31m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8054f7d8df5f4e909f0265e2b14f35e7 : SUCCESS in 1h 56m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/156e20b103004822b10a9d71607fed46 : SUCCESS in 1h 06m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/739a36165c48433d941f2abad96e6fad : SUCCESS in 23m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3bf8d41072ab4bde99ce52519532dd5b : SUCCESS in 5m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c1507c338e014c3f9509ea0031bafb47 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/38fd8d1072e84bcebfb9b63f8e9c4eeb : SUCCESS in 6m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3f5aead78b744cc2b88d24a08c9a106e : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2b3c5f16623f4cbeb4d1985ff1715119 : FAILURE in 38m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8cd7232701cf4a5daa86a2bfafee67ff : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5c60b7c592c84221835a558c7b37f884 : SUCCESS in 1h 55m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/539eae7d443c432dbbe63698a9456b8d : SUCCESS in 2h 35m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c1b4e544db144fb7a7f236448b74f51e : SUCCESS in 1h 58m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9b76fb4a90cd49c4861a93f3f48199ea : SUCCESS in 1h 42m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8e5af4b4fbbc4db6ba168995d93ad502 : SUCCESS in 40m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/16b6e6fda7204e698837cf957896dc7d : SUCCESS in 40m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ebc50a4c50c341b4a6f2b7f802c4c7fe : TIMED_OUT in 3h 07m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7bf725f2df4b4eddadd552aaf83542cf : SUCCESS in 38m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c8cbf92088f74119abd48707ba099ec6 : SUCCESS in 2h 13m 26s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/777840e1587542f48383fab898f994c9 : SUCCESS in 1h 03m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c7e28f64b6b04a21a0d16f5e426c36ac : SUCCESS in 1h 22m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f4dda6648c441008a94c02c3371fd1d : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":36},{"id":"df14f65bcbd32c7a38ad164c82475b39eb52a0db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-07 17:58:02.000000000","message":"Patch Set 37: Published edit on patch set 36.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"57da4463875450a305ee5fefb94e3e1da4543eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-07 17:58:55.000000000","message":"Patch Set 37:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 949006,70","accounts_in_message":[],"_revision_number":37},{"id":"9822beccb62b2fc8b960a002f0b217c54e6547ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-07 18:00:00.000000000","message":"Patch Set 38: Published edit on patch set 37.","accounts_in_message":[],"_revision_number":38},{"id":"73a04e7a3509fd2995b0a6d1cc21ac4b09859ad4","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-07 18:00:15.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"35b6f9fcd451403a1bca287a13a7a58838617222","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-07 18:00:55.000000000","message":"Patch Set 38:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 949006,70","accounts_in_message":[],"_revision_number":38},{"id":"cf4fda167723ef989ef996fb2b4b3d4af04bc81c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 18:41:39.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10f3f79ff8c54693ad199f2271a41e34\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cac3c85334f7430e8207d922f638b9d0 : SUCCESS in 12m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3f9225120aee4894a5d2372e929ee272 : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f9d9d7406bbd48509f79fbb1688485c6 : SUCCESS in 16m 04s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"a529e971140bfd4154ff6260f5b07bbe9abcbd49","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-07 18:56:58.000000000","message":"Patch Set 37:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":37},{"id":"76d2aadad11238bf81cb21fa3677d1d54dc1cadb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-07 19:00:06.000000000","message":"Patch Set 38:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":38},{"id":"4b9ba5cb4ced429bb1bc0154b9703792d0e9f5e9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-07 19:09:31.000000000","message":"Patch Set 34:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/125/ : ABORTED","accounts_in_message":[],"_revision_number":34},{"id":"0dc7e2199bf7fb0d5a089222b409d8902791cea8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-07 19:09:33.000000000","message":"Patch Set 35:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/126/ : ABORTED","accounts_in_message":[],"_revision_number":35},{"id":"e53add5bf25fa1dc821708339a3c2d8910a4f1b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-07 19:09:34.000000000","message":"Patch Set 36:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/127/ : ABORTED","accounts_in_message":[],"_revision_number":36},{"id":"a2105b18295eec7e019ec5a275c012462336858b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-07 19:09:45.000000000","message":"Patch Set 37:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/137/ : ABORTED","accounts_in_message":[],"_revision_number":37},{"id":"55720ed969a41f780a7f0f4ba4a1b8f018bd0b43","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-07 19:09:47.000000000","message":"Patch Set 38:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/138/ : ABORTED","accounts_in_message":[],"_revision_number":38},{"id":"8684d6555b98a57426141d2506d649b2ee3e652e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-07 19:09:54.000000000","message":"Patch Set 33:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/124/ : ABORTED","accounts_in_message":[],"_revision_number":33},{"id":"78f742b4c4d2bafcaee7b336e4e0270ef58007fe","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-07 20:43:40.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/57ee5fe9072940c890bcc17f3db68730 : FAILURE in 2h 16m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c60b2918a40a4f399fa1a922c229dd54 : RETRY_LIMIT in 27m 31s","accounts_in_message":[],"_revision_number":38},{"id":"56fff09847f24056b52c0f1b83c3951d35130678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 20:48:46.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1313a3c9f544c46a3c0e29d257ca717\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee24c5130e3f4d21a551053c39a24253 : SUCCESS in 8m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1339837a105f45db8cd4a517ba6ca15e : SUCCESS in 10m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b744705f56849ca8b83b332d60dff2c : SUCCESS in 12m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c675d730987644bda35f8a4833742d19 : SUCCESS in 13m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/7ede685e678b4afe82fe898ed0d47b42 : SUCCESS in 52m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fdf993d5bd8a4472926306d51ad5990a : SUCCESS in 1h 00m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/24c17b8c70fd4a1898dcc3a2b7105d0e : SUCCESS in 1h 42m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e353681330cc41b2bb10ee5cfa37392d : SUCCESS in 50m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/465b533d9a144da9ada04198193bd306 : SUCCESS in 26m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/57643e7b2dfc4232963dd7f69b5e8ac2 : SUCCESS in 6m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c54a408f7d484cf884bc39e0438bab04 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/34469e897b5e4baea2fcd9bbfee9d592 : SUCCESS in 7m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e715b377cb864d0c9ad463f6929834a2 : SUCCESS in 7m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b515a0492b364d9195c9bba006c778e9 : FAILURE in 32m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c239d83c1624817805292f810fccfc9 : SUCCESS in 7m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9903f731a19042168c894fd386daede7 : FAILURE in 1h 44m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e6252f63b4974d6a86bef844272baf8e : FAILURE in 2h 21m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/200bd4fdf5404318b210d385cd0a5fea : SUCCESS in 1h 31m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a50c7ae6dc494673b4a708ea77f699e8 : SUCCESS in 1h 51m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e959b11f697c42c4a1d69f3cb92ffca6 : SUCCESS in 31m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/243c9e7f08c3490b9698a39d0b538c63 : SUCCESS in 58m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/94f05ec51e7b41f6b8b180c71efcf5dc : SUCCESS in 1h 23m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/17767260718f4989aec3ba89f7b57b52 : SUCCESS in 50m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/770b48bf32f54408b825a10f3c85d147 : SUCCESS in 2h 13m 55s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/df4c53f6501048229e764bb23fe778a9 : SUCCESS in 1h 56m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/19571ffa9b7c4637b475cfd8c1ee2f1d : SUCCESS in 1h 24m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3d9ec6b011d4e8490094d5738bc9605 : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":38},{"id":"39f231cc5e119b1d074fb45c2d106f47c4bdeabe","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-07 21:54:35.000000000","message":"Patch Set 38:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 1h 18m 53s","accounts_in_message":[],"_revision_number":38},{"id":"a6ea2799570f3116632353cdfe311e6f7b3958cf","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-08 05:38:29.000000000","message":"Patch Set 38:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-38-fc : SUCCESS in 58m 40s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-38-iscsi : SUCCESS in 1h 5m 20s","accounts_in_message":[],"_revision_number":38},{"id":"f8f9087095f3b7c35743ce7213017dccb5b42c66","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-08 06:24:34.000000000","message":"Patch Set 38:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/653/","accounts_in_message":[],"_revision_number":38},{"id":"a7a24421d8872173094cbd7baa26c66cc9f9fbf1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-08 06:24:38.000000000","message":"Patch Set 38:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/38/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/38/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":38},{"id":"6b565d4eb350b8aa665a802ec996bc1712c7b0d2","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-08 17:19:58.000000000","message":"Patch Set 36:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-36-fc : SUCCESS in 59m 9s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-36-iscsi : SUCCESS in 1h 8m 24s","accounts_in_message":[],"_revision_number":36},{"id":"fcd84b227ae254fd230d827fd036ad2170d5ff0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-14 08:17:29.000000000","message":"Patch Set 39: Published edit on patch set 38.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"054e63692d1271d1bc30ad77fd9f73a70829f300","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-14 08:18:21.000000000","message":"Patch Set 39:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 951157,81","accounts_in_message":[],"_revision_number":39},{"id":"6c91add6d3a543bea28cbd0170c450ece8b7c992","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-14 08:18:23.000000000","message":"Patch Set 39:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 951157,81","accounts_in_message":[],"_revision_number":39},{"id":"62fe5551d1992f364909191d9664d527f9129285","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-14 09:16:31.000000000","message":"Patch Set 39:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":39},{"id":"3a895b102a8b1901bcbb2b789158ec38b696cd71","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-14 10:32:57.000000000","message":"Patch Set 39:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_205/index.html : SUCCESS in 77m","accounts_in_message":[],"_revision_number":39},{"id":"c0e52e0744fcbb9f6ed0865b1b99540c236c70d2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-14 10:41:18.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 1h 21m 45s","accounts_in_message":[],"_revision_number":39},{"id":"63dfa719636db45297d8a2e6cd935a57102815c8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-15 06:07:08.000000000","message":"Patch Set 39:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/aa981d27c1cd426ba89dd9efb76c75d5 : FAILURE in 1h 14m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/76071f79adef4a3b87f858b230eace1b : RETRY_LIMIT in 17m 00s","accounts_in_message":[],"_revision_number":39},{"id":"c5f9a4e49a72389d12efd5c9800c681fa2e14362","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-15 10:12:02.000000000","message":"Patch Set 39: Code-Review+1","accounts_in_message":[],"_revision_number":39},{"id":"c5c5482f214ff93fb182909cd85de5a98ce4d8f7","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-01-15 10:12:54.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"5d5acbf0f16ea579ff2fe01d9b81eb0b0c54ec82","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-15 10:13:36.000000000","message":"Patch Set 39:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 949006,65","accounts_in_message":[],"_revision_number":39},{"id":"45308e2d6914b2ca937942e3cdae49a2cf4ffb67","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-16 01:58:45.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-39-fc : SUCCESS in 1h 25m 36s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-39-iscsi : SUCCESS in 1h 21m","accounts_in_message":[],"_revision_number":39},{"id":"76601bb4a928c398121250cce8391db8ba2e721f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-01-16 03:33:51.000000000","message":"Patch Set 40: Patch Set 39 was rebased on behalf of \u003cGERRIT_ACCOUNT_37470\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"_revision_number":40},{"id":"3e70f0b437b09762b685ef4dd5e1017236fe163f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 03:56:21.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfa6ef9086ee4c9f93a8140385f3e53e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c613e8d9acfe4986aba7850df7067b69 : SUCCESS in 20m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb26849eec1949538fdeebab6910de33 : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d8fdfe4e208048b5ae36ead9e549682d : SUCCESS in 16m 41s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"b26d25e8e59211a59b9979a00dbe90438cfdd289","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-16 04:32:23.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/dc7e270fd64742d6b331c5d77675180b : FAILURE in 26m 24s","accounts_in_message":[],"_revision_number":40},{"id":"0ba00ce82854549f330c4fb1a1d30fba80ee8d1d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-16 04:33:46.000000000","message":"Patch Set 40:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":40},{"id":"3c8c45bcb30e094d254c71ab8d7186cc3835ca25","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-16 04:44:24.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/b798a9d802684adaab5ccb71a80a5aca : FAILURE in 31m 58s","accounts_in_message":[],"_revision_number":40},{"id":"8002610dea0aa15586115c839a34f6adb7364dad","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-16 04:51:42.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 35m 31s","accounts_in_message":[],"_revision_number":40},{"id":"72d2a87fc13fc122b68a1cf2a9a67fb68e294948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 05:40:36.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e05413e033224122a831148f837f0f06\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe023cb59994489d8acd09183a8c58f1 : SUCCESS in 7m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5590043e2efc4bcdae1edce044bdf98a : SUCCESS in 11m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d36a1507c8954efea35de4bff03219de : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdce8cafb40f4cc5b410a2f356a9a8fb : SUCCESS in 13m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/16e3891266484a60a6a16ef0a92f0300 : SUCCESS in 59m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/624732bbdeb946a5b5fb6f70149f4dc8 : SUCCESS in 1h 01m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8f11803a6aa648e7861e7072bf3c9c8e : SUCCESS in 1h 28m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39b8aeca1191441f8de2cfa8e84579c9 : SUCCESS in 1h 06m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c14deb3f37c34b1284da694cb7b61d03 : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8090ef5a33fe4e9aa049af409d41ebb4 : SUCCESS in 3m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea91a9a5233143958ad8f500a0bf3de0 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a11bfcb32f3c478bb6ad292f3c19da9e : SUCCESS in 5m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5eebd2c0785349db99fdeafc8ff6cf97 : SUCCESS in 4m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a4c6df537aa74a90aeebd6b6e9b7df8e : FAILURE in 33m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f5083f3d82a2446c8db017fa2a432b37 : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4ae2c4007b34420b81e24f67fa0d8ebb : SUCCESS in 1h 04m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/346698529c154f85809b7efc6038cb48 : FAILURE in 1h 27m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/06fd517730d54a0f9721db88b0837103 : SUCCESS in 1h 49m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2f1c3c2cd6214650beb206af4a5af210 : SUCCESS in 1h 57m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5ed10fe06edf4f2f80344e7aad50b01e : SUCCESS in 31m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/abd9a059b6054abeae01ddd44ed240be : SUCCESS in 1h 13m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5df85d2110894b8b900ea09c8d6e5e5b : SUCCESS in 1h 46m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/077b5afa46a04dac915ed5411d9c4f7e : SUCCESS in 26m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2a0092330d31454fbdaeb28de73e09ed : SUCCESS in 1h 56m 14s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/efd1cbfbae514b8c933aeff5868f5fb5 : SUCCESS in 1h 11m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8732e8c05a9048c0ae56e0cbb8c1b2c5 : SUCCESS in 57m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49b79bb6cab34dd98bb380a1a04aa444 : SUCCESS in 49m 18s","accounts_in_message":[],"_revision_number":40},{"id":"0f2b6244c7f9e59db7d89127cbe9ac583b9416fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-16 05:47:38.000000000","message":"Patch Set 40:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_01_2026_05_40_46\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":40},{"id":"45efb6f9c4b6f326aa98f7e0a35e91ef796d33e7","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-16 06:16:19.000000000","message":"Patch Set 40:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_235/index.html : SUCCESS in 80m","accounts_in_message":[],"_revision_number":40},{"id":"568f804c6d3429dbd81f4bd3b89493772e733287","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-16 12:24:58.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3bee5d0b7a0c4f9e917315ca26cd736d : FAILURE in 1h 14m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e0d76bda5c6c48328f8fca441d5a75a5 : RETRY_LIMIT in 15m 32s","accounts_in_message":[],"_revision_number":40},{"id":"1906767fb9e4810352498164146a4186d320def6","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-01-16 14:25:33.000000000","message":"Patch Set 40:\n\n(3 comments)","accounts_in_message":[],"_revision_number":40},{"id":"67b018e9017aa5da7542e09f797d369148a1df19","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-01-16 15:18:41.000000000","message":"Patch Set 40: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":40},{"id":"37fd009b5227b27d2edeb937a1a4e2ff5cc7240c","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-01-16 15:20:35.000000000","message":"Patch Set 40: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"52c74d152eecdfe1a95fb73b799cf91a8d113665","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-18 23:42:18.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-40-fc : SUCCESS in 1h 35m 8s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-40-iscsi : FAILURE in 1h 31m 45s","accounts_in_message":[],"_revision_number":40},{"id":"62059454c76b54e80880032bdc54d07a304f8e90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-01-20 05:52:50.000000000","message":"Uploaded patch set 41: New patch set was added with same tree, parent tree, and commit message as Patch Set 40.\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":41},{"id":"ae4b55e66f643464be1f0be49a6de2c0981a880a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 06:46:40.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef6b024f509946cebc0ae17f08746c31\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3c56fede76f4f66a44a86b8a8559918 : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1866d743783b43e6b9dc5c67559a4c71 : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0656d732972b42958e6330530f05bda5 : RETRY_LIMIT in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"c2bdd16ea8797ff73a21d2c2d882f462c6e9d3ef","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-20 07:04:06.000000000","message":"Patch Set 41:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":41},{"id":"71e3cd6ef8c64aad55eac870dde750c18ccb1491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:06:10.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7644330b9e524e2da10c7da4bde20cc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b605f6aa689a4cc8b885888546e12953 : SUCCESS in 7m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f42dde1cf5684ea49ad84cdd45b7a64a : SUCCESS in 13m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84feb26340a2480ea973a2989d555af2 : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf15f0664fd8438f98d41cf8055aa155 : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/ab760b1773414730a801f1d11426a616 : SUCCESS in 44m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c9d489cda6e41f19849ca6677c9cd86 : SUCCESS in 59m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/96a5344c296b4feaa1f5628d15fa74bd : SUCCESS in 1h 24m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc2a1dcadbe24e17be42f8f6034b7b8a : SUCCESS in 1h 04m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/48596e6066b147dfb1ceb430c01e9f04 : FAILURE in 25m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5a298722330f40a888b3b00e29b8bbb9 : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7229ac21de3a4d8193a6ce5cbb91f30a : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b260e25c12964d2187c51f4276b28e17 : SUCCESS in 6m 00s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/329f2c2fe4194ff2a24b13346ae814b0 : SUCCESS in 3m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7a80e1e90bc942b9849d00bc1d2f2cd8 : FAILURE in 31m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/80c50dedcb0c42bf87955d16ac801254 : SUCCESS in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/374b7f55b9c04bbf8f202ade32a23d20 : SUCCESS in 1h 54m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0d1143366d0e454a89e2625387690f07 : FAILURE in 2h 55m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b5c4c34f00ee4a73a64514d57c3a2508 : SUCCESS in 1h 55m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b230645806fe48f39d4eb5ca2d52b8f8 : SUCCESS in 2h 10m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b72a684289b347358bfdeb50927bbaeb : SUCCESS in 38m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a7e1cdb10cd540b1b42face0873faf5b : SUCCESS in 57m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2874e7d06c84455880ee533106d125bf : SUCCESS in 1h 37m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5cb1c446ff1a43269316aa7f1ffae0ab : SUCCESS in 31m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/772d9ecee8e14fb5b4a7e831225375cb : SUCCESS in 2h 13m 30s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f7f3fd3f209f4e9ba50691ea9a21e120 : FAILURE in 2h 28m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0e47812455834239a501e895a4c29665 : SUCCESS in 1h 53m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2912369861f04e1ea14885a8281aca65 : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":41},{"id":"4fbbff58ecc792bf63eb38429055884cfea723bd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-20 09:14:10.000000000","message":"Patch Set 41:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_01_2026_09_06_21\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":41},{"id":"bc2fde73adccc95b310632f252f3ebc8f30ef55a","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-20 09:54:57.000000000","message":"Patch Set 41:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/75719acf72174d50a7d3256534484f49 : FAILURE in 27m 18s","accounts_in_message":[],"_revision_number":41},{"id":"4226a1472f24da3045c57dda348a9314eecccff0","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-20 10:47:56.000000000","message":"Patch Set 41:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/63d89db7d9e6427d9fe1cfcb0dc6739d : FAILURE in 35m 25s","accounts_in_message":[],"_revision_number":41},{"id":"4d3ecac9ca21c6a2ef957b3a3881c738f285639e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-20 13:32:56.000000000","message":"Patch Set 41:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 36m 50s","accounts_in_message":[],"_revision_number":41},{"id":"ee5fa02382940648e3e09a5f7c402c552eb9578a","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-20 14:12:31.000000000","message":"Patch Set 41:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_301/index.html : SUCCESS in 98m","accounts_in_message":[],"_revision_number":41},{"id":"d12af9163a74da805169aa15cb276fd691a95568","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-20 16:31:46.000000000","message":"Patch Set 41:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ce53ec7a5cf84e819e1b9d7bcf67901a : FAILURE in 57m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e336fbc6c89c47fc8d1c703fb2e52ae0 : RETRY_LIMIT in 15m 24s","accounts_in_message":[],"_revision_number":41},{"id":"a3811c3c0c2b0879d06d9a343ddbce2134470a3d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-27 15:52:23.000000000","message":"Patch Set 41:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-41-fc : FAILURE in 16m 50s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-41-iscsi : SUCCESS in 2h 56m 41s","accounts_in_message":[],"_revision_number":41},{"id":"53d7800bdf736b90ac76608b561cb9e24c81fcc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-01-28 06:59:48.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"26482ef88237f005f00c75584f296c8a341824dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-28 07:00:50.000000000","message":"Patch Set 42:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/687/","accounts_in_message":[],"_revision_number":42},{"id":"17f74889ede2e01487341b65d93221a34f15a18a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-28 07:00:52.000000000","message":"Patch Set 42:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/42/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/42/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":42},{"id":"cabd6063e89c3a37d985d5543efb68cfa3b62d78","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-28 08:02:49.000000000","message":"Patch Set 42:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":42},{"id":"bc11423fd991d279c8d5cf681f973c07ba496c56","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-28 09:34:37.000000000","message":"Patch Set 42:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/951157, https://review.opendev.org/926844, https://review.opendev.org/949006, https://review.opendev.org/929906, https://review.opendev.org/938771, https://review.opendev.org/916722 are needed.","accounts_in_message":[],"_revision_number":42},{"id":"b63b1e01bbda260d7801e4b4c10d8b83bd4020c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:09:33.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa8e8b829ce540a5a23e5d435f708e2e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd3d49b961484c33b79a928da337ee4f : SUCCESS in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d53aaef32ed4db89a000143f4dad358 : SUCCESS in 11m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b96a7b48497846eb82c6968c268631af : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e6b0434dc141f389e248848901f11b : SUCCESS in 15m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/12c3a385c8864afe8f630c0dc061de72 : SUCCESS in 1h 01m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/086d7b0c2939425ba3346f9deebbca09 : SUCCESS in 45m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1595cbf94c7942849cf928cc6898ca6d : SUCCESS in 1h 39m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d92b42da99b4d9e9e93862fadd461ea : SUCCESS in 47m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/36f1ddc1f38e41aba5afb23ec7a2a6f7 : SUCCESS in 21m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0832c7569955405c99ca74d6a61ea252 : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9a57771d29504f278a4eea75a22478c0 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/31adc8120a8e43d0bc79fc8ed5cc87fc : SUCCESS in 5m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/57b0300e9409477e9a3fd7faf03342a7 : SUCCESS in 8m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9d8bbc7f0d524cd99473cdc4b9510552 : FAILURE in 44m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eebef6fbda2b4150a2f3a5ec245bf89f : SUCCESS in 7m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/11dbf16543f145fea19db132c8ed60f4 : SUCCESS in 1h 49m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/889bde1a7fb8496ea1dfc0ffd6d32338 : TIMED_OUT in 3h 05m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0bf5f0dc7c9b4a1980d8c24c4f78feea : SUCCESS in 1h 56m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/510385d2904e4f1b9891403174f4e459 : SUCCESS in 2h 14m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b24484207ae24ba3b73bed173161eeb7 : SUCCESS in 35m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a68d655a0a5c4cc99b3e174772ec6db2 : SUCCESS in 1h 16m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0eff97e5a1fd45e6a240c4ebfba6093c : SUCCESS in 1h 38m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ac8d4b25fa1341c8a3c4ed56cbd395d2 : SUCCESS in 34m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5a1de71f2f4740e99b3b998f843e899c : SUCCESS in 2h 42m 39s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/b5f3c5a79c5b4f1eafcd29756977a8b7 : SUCCESS in 1h 49m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/19c101510e624975b41e7e6372428eed : SUCCESS in 2h 03m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8e4b3bbef564ff4901721d38cb29a2d : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":42},{"id":"70dd9dd042f7963cba893a775b33d47035ae2e2a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-28 11:18:42.000000000","message":"Patch Set 42:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 35m 24s","accounts_in_message":[],"_revision_number":42},{"id":"ec75586260faea0b7cd34e29d3069962452df78f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-28 11:40:29.000000000","message":"Patch Set 42:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/353914a13d44453aba5ede7e01cc49b3 : SUCCESS in 43m 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a336b053dc1b4954a5b32115ae0d8dc9 : RETRY_LIMIT in 15m 17s","accounts_in_message":[],"_revision_number":42},{"id":"caf6ca72bc8f5cb6b4754fafab3fae31a2fee2fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 11:46:05.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/590600e1e302476f8252efefa44883fc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0eb1247f9dff49be9bcd957f0e30adcd : SUCCESS in 21m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8b040b730f454cf997922595de68276b : SUCCESS in 23m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/38fc39e4d9f54675983534b5ae3c8f7d : RETRY_LIMIT in 6m 28s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"47c6f34c80e4f969365bb45ad1c66ee6dff1020a","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-28 17:58:09.000000000","message":"Patch Set 42:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_378/index.html : SUCCESS in 94m","accounts_in_message":[],"_revision_number":42},{"id":"03fa64b8df3446d030c2cb9160c68a36bd21baf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-01-29 11:51:22.000000000","message":"Uploaded patch set 43: Patch Set 42 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":43},{"id":"61708fba026a398bad11e0353ea0dbb3c7a658e9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 11:52:32.000000000","message":"Patch Set 43:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/731/","accounts_in_message":[],"_revision_number":43},{"id":"7072a6f86c619f03117f1fd70a8241e4ccf33848","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 11:52:36.000000000","message":"Patch Set 43:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/43/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958465/43/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":43},{"id":"6b47a750a75042f6cc919acc3bf47ef046ae449e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-29 14:47:21.000000000","message":"Patch Set 43:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":43},{"id":"63e43e90be1fbaeeb6e81f43ba0f206785dc6bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 14:58:20.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3941fdea0d3f410f990df30568a01a31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c527fd73919c4328af5fd9886b1731ae : SUCCESS in 8m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8d6da590b764f7eaeda000173c013b2 : SUCCESS in 11m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6bb83767a534597b8f5db4fc850dc6e : SUCCESS in 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73d7fb71cd3a4c898f53299e6ef93042 : SUCCESS in 11m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/75a27e7ff902414d88a60abd87a4c1fd : SUCCESS in 57m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9806b06697942ad8bc62aed2037c341 : SUCCESS in 59m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5c3a3624c8549b2b268914795292310 : FAILURE in 1h 30m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/018b7e4a8d4144d1bd1e9c483754bdd7 : SUCCESS in 56m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/131801ed4acb41918325bcb417cddde3 : FAILURE in 25m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/207ec1cd7b2c4bb2bb6528f2c5a2d171 : SUCCESS in 3m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e19d12213c0048a1b8e3b701b84c7bf6 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eaa4708d2cdb49b39346be10174bb2c9 : SUCCESS in 9m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/afd9f15f40054b8fb634cb7c5565e604 : SUCCESS in 9m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ff2b5c29a647432fa50becea473b5e0d : FAILURE in 34m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0066c819ed02430c80a51f811bfed702 : SUCCESS in 8m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/94672df0396f4a97a797acced88a139a : SUCCESS in 2h 11m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/498a144f6485477c9d42587bd764908d : FAILURE in 2h 55m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a1b779ac26254bf387816795db05560c : SUCCESS in 1h 42m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/263680653833428e9347a6c50aa350fc : SUCCESS in 1h 59m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1643042fee4f4a1a9c2bf2a555198ab3 : SUCCESS in 22m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/27e18aea32024e77ae1b046902cfbb5a : SUCCESS in 1h 11m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5c7281d438a74562809c87c8ea70d777 : SUCCESS in 1h 43m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1b691af3c9084d87ad6b8cf2cd7c32e8 : SUCCESS in 43m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a7af47523ab948939ebb9d920d8f831a : SUCCESS in 2h 01m 13s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/47448e4d893140a5bb2c0e15401e4e2e : SUCCESS in 1h 56m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b7bd28872fba4fb9a276f51a8650b8a9 : SUCCESS in 1h 52m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/939fbc5e0f60460892adf431adf5e12c : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":43},{"id":"0bdf5627bf7278b825bf93f0081b8a14a6cf68ec","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-29 16:05:25.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 40m 37s","accounts_in_message":[],"_revision_number":43},{"id":"92b49e9d5f6dee6cf991e9be206cdba779509709","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-29 17:08:06.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/d7eb6dd071194bd4badbabd2ee788381 : FAILURE in 34m 51s","accounts_in_message":[],"_revision_number":43},{"id":"4a7500d69dbac650aa49670e54c9a215667e5977","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-29 19:22:44.000000000","message":"Patch Set 43:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5f57d106352645fcb1581cc49aac1c9a : SUCCESS in 42m 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1820ff2a85484c94b4229e6c31175375 : RETRY_LIMIT in 15m 23s","accounts_in_message":[],"_revision_number":43},{"id":"1b93c105943cedd53e83641df152067b0d592281","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 22:17:46.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/705e3de7c27641d5a86e47f1f102f70b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6ee019838de43228c575ffef8c42f10 : SUCCESS in 18m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0eaae2c258a14969ad18a09c60a943e1 : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/99d11ac0025141b590c2b67ce03dd671 : RETRY_LIMIT in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"5c1fba17186b53bc15683f565b43d7e53c9da866","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-31 22:23:17.000000000","message":"Patch Set 43:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_422/index.html : SUCCESS in 100m","accounts_in_message":[],"_revision_number":43},{"id":"0125a76dce3930e6632c1a50ba71699112b1c9fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-03 08:44:17.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":44},{"id":"8224954b417fd3b49fc3677e731fd2c1c45fd49d","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-03 08:45:16.000000000","message":"Patch Set 44:\n\n(3 comments)","accounts_in_message":[],"_revision_number":44},{"id":"5d1b4dbdc4b01692d274b3cbedf5ece508663935","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-03 08:45:41.000000000","message":"Patch Set 44:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":44},{"id":"c79879b824f0cb9e48723b24fbc60f438e758a6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 09:28:26.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d728dcad76734d819d0a1ad909bcd028\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/77486fabbf43480dab0922a49e0aab08 : SUCCESS in 20m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a30722bf77bc4320853c36f77e7e5e91 : SUCCESS in 21m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d10b203d5aab49b3ade1e9d0de73c330 : RETRY_LIMIT in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"64edf892f68aad6535fe614fa945de96956eb76a","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-03 09:37:39.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2bb9ec36ac9f47b4bc7cc4b52bd3ef5d : FAILURE in 34m 35s","accounts_in_message":[],"_revision_number":44},{"id":"56d1a791f4b80cf0ca8f7c86c723a22056c00a51","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-03 09:38:53.000000000","message":"Patch Set 44:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":44},{"id":"9a93d0654a648377a59972ce79801f81afb68a75","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-03 09:41:07.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 36m 47s","accounts_in_message":[],"_revision_number":44},{"id":"0b9de44ac0a503f209043c1ebcb7589284e771eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 12:09:46.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20d207bd96c149fba7a59ded57ab8165\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b879185647694293875fe19cc3d564a9 : SUCCESS in 8m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/690ef0cd18814a96a5b3b035fea97d37 : SUCCESS in 10m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67a4241469fa4f2abfef134ba60bd7dd : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820d2255b40c4fdba1e2f09bf6a9c929 : SUCCESS in 14m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/d5dad23d0e9a4d1f8d323fdc83275eca : SUCCESS in 59m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/804b7e74027d407899d0759022d9242d : SUCCESS in 1h 01m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f09925afe176479e9c2e5c8929e20871 : SUCCESS in 1h 32m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e85884280ff4d38b451d70f72f0cd01 : SUCCESS in 1h 02m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/03817bbd3be24d1f9b89bc105f0e187a : TIMED_OUT in 41m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1a6f1f6d66cc42ce92605903a11dd8ca : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5fb1521aa90245aa9e11e714deda8ee3 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/af9daf07d11d4b25aa554a8fe98d534e : SUCCESS in 6m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/866735128b3d47dc99487665e0ca0aaf : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5d69cd0309e74a33acf07d48fd57a55b : FAILURE in 38m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/45dbee01f405495cbea4a8affb488dc1 : SUCCESS in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0efc94a3d6f44089b7070778eca2eb77 : SUCCESS in 1h 47m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3647f5c9e2724166a5c56fd0ec6bf20f : FAILURE in 3h 01m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e6c9b641a50481cbd8b7abc8f8e3796 : SUCCESS in 2h 03m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d1c06c0d1e674ea4aeb64b8f672c9869 : SUCCESS in 2h 16m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e43b98c748084df3bd7466f19c2fc052 : SUCCESS in 25m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/235cc02e74af44a2a367e6ccf8adc4ab : SUCCESS in 1h 08m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/44ff3430b58544bb9cd5242a20c682a0 : SUCCESS in 1h 13m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38b2668396064f86ad64c658209959ea : SUCCESS in 25m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/80930197798746e6a5889bf999d5c957 : SUCCESS in 1h 10m 32s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/32da557c9f544217b4423c838d6dd266 : SUCCESS in 2h 02m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/97464f3c1e0a4333b5dcfd9a5e9e8ba5 : SUCCESS in 1h 44m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20aa199c5d584a90b904c14f296a8014 : SUCCESS in 1h 07m 23s","accounts_in_message":[],"_revision_number":44},{"id":"39297996ecd7ddf800585e33be7098c78f4151d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-03 12:16:47.000000000","message":"Patch Set 44:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_02_2026_12_09_57\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":44},{"id":"435173cb6df300e6ef67005395ef90752b1ae0f8","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-04 01:10:24.000000000","message":"Patch Set 44:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_483/index.html : SUCCESS in 98m","accounts_in_message":[],"_revision_number":44},{"id":"fdba1ae37bd49f0b6d1f807696f9ca339c6b0fc7","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-02-04 15:05:53.000000000","message":"Patch Set 44: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"a877a48efb6525318aa78294791c80c5c9cca992","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-05 06:38:56.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":45},{"id":"34e3639d1e6cc67c0af3986350d94cd50f5341d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-02-05 06:52:25.000000000","message":"Uploaded patch set 46.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":46},{"id":"b9a96757b0816c945535024a3a80da9bdc8a40f5","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-05 07:19:48.000000000","message":"Patch Set 46:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 975638,4","accounts_in_message":[],"_revision_number":46},{"id":"419f9f29c547955f19430439f7c6962e768b8c72","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-05 07:19:49.000000000","message":"Patch Set 46:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":46},{"id":"aabdb808724153c055ee00cd92e3cece81dee451","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-05 07:19:56.000000000","message":"Patch Set 46:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 975638,4","accounts_in_message":[],"_revision_number":46},{"id":"4fa64143b8d6a6c988528ac653fd7cb28831f07a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 08:41:03.000000000","message":"Patch Set 45:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":45},{"id":"6e79d6b9ceed204131fe40ff226421eaa98ad46c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 08:46:52.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbd44841d8c44cf2a4f583adc3e0a463\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/160f8677895d4490a7b7d33cf1c53873 : SUCCESS in 19m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2fc0527411143ec913326372cef808c : SUCCESS in 19m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/659890a09b97476bb89c496728400fdc : RETRY_LIMIT in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"d62200099b75c224f47eb250c44aa514087f2480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 09:31:25.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dda849ea569485095be2e99965ce423\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b3a3c41520453cb9725874c5b2544d : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5535218db3414c88805015d8beca0764 : SUCCESS in 11m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebdd589fe4524525aed07fcc4e6b8210 : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1c660b36bf4b6eaa53f85a01439a61 : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/dfbc9af1bc7442c98ee5d24cf8966b7b : SUCCESS in 1h 01m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1290f25b25f44f6894c4f376ed7d15f : SUCCESS in 52m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13adf9c883674801a991b796b198ad99 : SUCCESS in 1h 27m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c26a09c9c3f48bda6e46b89b5fe95f2 : SUCCESS in 34m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3449077b57d64cc6bc93d6059af1437a : FAILURE in 23m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/754caa4032064db5af7cf2144cbb2062 : SUCCESS in 7m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d77dd6fa410a499fa7a46c5fd872124d : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/99c23231b849423dac0dcdfaf8373314 : SUCCESS in 5m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f0c06591139f45279ae6208f48af9dc2 : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fb56976c30724e6d851a8e168f7e441d : FAILURE in 42m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e8be3c689a4481ebccdb59fac4f4f41 : SUCCESS in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/88438a798e5346e793d1c1579b7615e7 : SUCCESS in 1h 10m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ac5f2b34c06b4e6f8fcee0f975832d98 : FAILURE in 2h 31m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0268a3fc9e0d4de38229b614de2f8b2d : SUCCESS in 1h 45m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3a5d55117c77483da0df9ff84c334c20 : SUCCESS in 1h 53m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/de8f3bc83ee540eba2e8d1e894faf1bd : SUCCESS in 22m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/610e5691308d4e428c0429a4e1ce9ab3 : SUCCESS in 1h 02m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/29781dd6a3804d7f810d014e37aa3867 : SUCCESS in 1h 54m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ec00ab56dd724511be06f20034081ef9 : SUCCESS in 48m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ba1bdd891c0f451cb673af3bf9f9840a : SUCCESS in 1h 04m 45s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/55a14d3fdf7144aab2f5033a4c785a6e : FAILURE in 2h 01m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/84cd0e6c43c14fa7acfaa4e274777fe7 : SUCCESS in 1h 48m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1fc0557578f46efa530f2241cedd47c : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":46},{"id":"41c0c27a564fcf15495089ce351d68cd3d930c68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-02-05 09:44:17.000000000","message":"Uploaded patch set 47: Patch Set 46 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":47},{"id":"cc2c85842dac10e5c23b5a68f9714c4e8ec61176","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-05 09:55:20.000000000","message":"Patch Set 47:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":47},{"id":"da0213a02b796fa914a0a6510feca58391a6ce1c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-05 10:19:25.000000000","message":"Patch Set 46:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_02_2026_10_12_23\nLeave a comment with","accounts_in_message":[],"_revision_number":46},{"id":"bf827f207a5c968f451bac762b0528accffb40fb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 10:54:45.000000000","message":"Patch Set 46:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":46},{"id":"daa1f64a9c172b90d9508bf38f881a54ddbb9cf5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 11:24:19.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74055752c3c74178a11bb301318d9163\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/80aa0cf079a3464492f3500e19da5055 : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/97614213416b49d8ae8a07a08e9f30b9 : SUCCESS in 13m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b1050d983acc4b3889996aaa277f890e : RETRY_LIMIT in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"3aaa1222a45c8698d85aba94bf0ede7f48c3603b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 12:45:32.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8a5b840267d44fa965cb07c3239b861\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09e7581ec12c436fb50ba61670f9203a : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88ee8cc3bf8b4a6a94211abb0a9d4292 : SUCCESS in 9m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5232b12820e845e2ac7b201b9fa4043b : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4927f0ae821940fb89b12ec4f5e944bc : SUCCESS in 15m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/cfb6d97abd88456da5ba912435158622 : SUCCESS in 1h 06m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/93f59e2375994dfca413f5a4a03af3a3 : SUCCESS in 1h 04m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ce61bf11b8746fb8cbb5c3b966aad7a : SUCCESS in 1h 21m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f24ab3bf8fd4f6d92b7f2d32623d077 : SUCCESS in 1h 06m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f8aab130dc124c9f8965b51978f14bec : TIMED_OUT in 41m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9f0547471289460583c05b40df8496e4 : SUCCESS in 3m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e7426af26b7435ea991ccafd4e38e9b : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28f411db961d46c793990554b5645892 : SUCCESS in 7m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1521ef77b03e4cf3adb176ceb8ef68eb : SUCCESS in 6m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/acb99636344e40fdb1a321e6afc29288 : FAILURE in 21m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a50dfc4a74ac4abe9b3a4cded290594f : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/54f93030567b49f3aa6343361add57cb : SUCCESS in 1h 45m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/14aafdcfb90e41f89ad09aca6b4e4db4 : FAILURE in 1h 31m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dcaba3c4bed14cd1994b971a4ae7b89d : SUCCESS in 1h 48m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d526c691e42e4f7087dae650287a7512 : SUCCESS in 2h 37m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a71cd4df1edf4997ba0ed2b5ac6b903b : SUCCESS in 21m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/60667e7315904d6c816b10cc497d9f9d : SUCCESS in 1h 03m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/82126a46a5964f6eab57436e65af50e5 : SUCCESS in 1h 36m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/72efb527d4bd4e548f1bc169b2409d28 : SUCCESS in 38m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/37dccd4ea30e476194d7ff24781f7276 : SUCCESS in 2h 24m 31s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f02823f66a17439aa4e4b164649e43a0 : SUCCESS in 1h 58m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/48601ffa660742fbac2a661589ede487 : SUCCESS in 1h 40m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ae2b0b13c84449e86cfbcb519476136 : SUCCESS in 49m 01s","accounts_in_message":[],"_revision_number":47},{"id":"f07704905a2ea3816ba0fd9143adc5dc9f72b346","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-05 13:14:31.000000000","message":"Patch Set 47:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_02_2026_13_07_39\nLeave a comment with","accounts_in_message":[],"_revision_number":47},{"id":"3df4e4a26daafac923df678711d88199cbcfe336","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 14:08:00.000000000","message":"Patch Set 47:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":47},{"id":"6a68767a4675d75b48807af2dc6afa5d579acf71","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-05 14:40:12.000000000","message":"Patch Set 47:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/8700e5289ec94e8aaed3b6981c9553a3 : FAILURE in 33m 18s","accounts_in_message":[],"_revision_number":47},{"id":"91ee9b08ba8fbfbf87919c1859e23dbaa89a54bd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-05 16:08:54.000000000","message":"Patch Set 47:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 36m 14s","accounts_in_message":[],"_revision_number":47},{"id":"4cab3d7d22384a88ed37c041e32f9fe971adc9b4","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-06 15:04:24.000000000","message":"Patch Set 47:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 938771,101","accounts_in_message":[],"_revision_number":47},{"id":"356d57dd0873798e686e6edb772589e6cbf0f607","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-02-06 15:13:45.000000000","message":"Patch Set 47: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"1cc5aeceed080f4b7ae8fdf9afe80e8967bbb220","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-02-10 18:42:21.000000000","message":"Patch Set 47: Code-Review+1","accounts_in_message":[],"_revision_number":47},{"id":"c0692214f02596916ebd41c1bde9940fc6f2f902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-02-13 09:22:24.000000000","message":"Uploaded patch set 48: Patch Set 47 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":48},{"id":"390cf702a1662bba964d7a2e3c0117a0589503d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:00:19.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9637f731ab934aceb118288d8dcb865b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7dca053c886046d2a0c49120a6106e2a : FAILURE in 13m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c61e486ec1543b68081b979808913ba : FAILURE in 9m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c7d5a939dcf412c98165ed544cbfded : RETRY_LIMIT in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"234518f249e638678ede9ad4983af3f1f73e7f17","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-13 10:02:10.000000000","message":"Patch Set 48:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":48},{"id":"c6b1231f675719e3baf62cb4356530bdc9353ff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 12:15:51.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e71d47c2c2454ddba283dcb1d6575667\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfae83a45eed4b6ea507d4ad47b5a1ad : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9a842f4dbe14ad781cd1b362ef6921d : FAILURE in 11m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/253f40d86d934d78ba9f4a18ba79044b : FAILURE in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dadbbfee41fe4320a7415c5b7f4d8dba : FAILURE in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/5c76a23c016546bdbda0696aa2b53d35 : SUCCESS in 56m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6571d489a9c146e58616f39e3b79dba8 : SUCCESS in 1h 08m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f361cd47900c48af9c2a37cf2f17b983 : SUCCESS in 1h 27m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5a9ba8a1bbc41a49bcf60eadb17794f : SUCCESS in 1h 04m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/18fca22c8c7a457c99ed390dcada4bc6 : FAILURE in 6m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7ca949ba6b974fda857704bdc5e4cdeb : SUCCESS in 6m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8c82d2922e664820ab5d65f55a78aa0f : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e715777d51a54c329a15d5f4ec6e5941 : SUCCESS in 6m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b5ef1011f9c14815ae89d227db793623 : SUCCESS in 4m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d9bdb4fe76b14e7ea4e9dfb8e3e4e935 : FAILURE in 23m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/80c00c7ff3594ddc8aee444b07280a00 : SUCCESS in 7m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fb05ee4e269a4123b06b922de6e97c85 : SUCCESS in 1h 46m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/28c0338448e0494e8caaa228cdef8246 : SUCCESS in 2h 40m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dcf263d7ab58483a97d8df155f679912 : SUCCESS in 1h 46m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f21165bdb2924bf299e98d9abbd90341 : SUCCESS in 1h 51m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1e7cbba69cf74ca4a9bfbdaef2663a0a : SUCCESS in 32m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/80d7f058a8634f4c876394b194c842a5 : SUCCESS in 58m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/626fec85711f4750a164c87babc0f289 : SUCCESS in 1h 27m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4217bd4cab7f4c9481ade9124c691843 : SUCCESS in 37m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/36bce4dc6fb6460ca0ef638172440d5a : SUCCESS in 59m 15s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d18c5203082a4c829b359c492eb9cfa3 : SUCCESS in 1h 53m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8cc3da66db934bbdbb7ccaded3f24925 : SUCCESS in 1h 46m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddbe64e449ca40649d90b1b0dd17ce92 : SUCCESS in 56m 49s","accounts_in_message":[],"_revision_number":48},{"id":"3e18e5a9b8b4c2abdce68a18b4a88ad653739d27","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 12:37:58.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/cc89d2f1744f475cb369d2d91548cdd7 : FAILURE in 23m 28s","accounts_in_message":[],"_revision_number":48},{"id":"22be5a8433ba7d37730f8afcc67b67ba45231a5c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-13 14:09:12.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 35m 26s","accounts_in_message":[],"_revision_number":48},{"id":"64297e47e0caabbaf0a844caf738c0e31cad8e42","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-13 14:18:02.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/3b28471b77e444e9a5ded3206e4610e3 : FAILURE in 34m 40s","accounts_in_message":[],"_revision_number":48},{"id":"9536ecba83745df9823f29c9c0ead7b2fbcc8a37","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-13 22:20:21.000000000","message":"Patch Set 48:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_642/index.html : SUCCESS in 95m","accounts_in_message":[],"_revision_number":48},{"id":"e362117f6e0d96ca17486bb9b171c3e676f1255f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-14 08:37:16.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c995c7482c18462c974e2f2c5216b19c : FAILURE in 1h 01m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c4113df320e34e28ad4c6527a184c858 : FAILURE in 1h 01m 46s","accounts_in_message":[],"_revision_number":48},{"id":"bee39d3e7c47934997b1c7b8551ac51e3644d2ea","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-02-19 13:00:27.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"2b0539d4b57fc1c6052fe587d3f984f894ff68ae","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-19 13:07:01.000000000","message":"Patch Set 48:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8c53e1abcd1b4e3b90003edf89c459fb : RETRY_LIMIT in 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8a9e2e801ac94edaa19fb3b89b391fa9 : RETRY_LIMIT in 45s","accounts_in_message":[],"_revision_number":48},{"id":"c9e5ac66c0e9c5b20a0b0da4ef11975b40f7656f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 13:46:41.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cd56b8463bd4e8d89fa42071b53338c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c78b42f4c471423d9acff3c3768b5011 : POST_FAILURE in 17m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3be96f06a723436eaf6f4c5f767fdfb4 : FAILURE in 15m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c9a2a021cd85476bbaece621eee27f8e : RETRY_LIMIT in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"37a796c702cc9beb5d72de5b1a8095e3a858ab23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 15:34:49.000000000","message":"Patch Set 48:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/268560be0a8b451f9e50500e71e7d0a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b2a2e5641134a7bb8299394df05ad96 : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7275c5d3715e47a5bdc82cedc987122e : POST_FAILURE in 11m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7dfcd5dfbfc49afa942fab8baf314a4 : FAILURE in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/914c7925842e480680652bcc6d735bc6 : SUCCESS in 14m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/6c16f0a0bdbd4bbfb98eeb72caf621d2 : POST_FAILURE in 59m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0459b2d5f8a741bcb391a4ff46f60f51 : POST_FAILURE in 59m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5db26d365dd467e97bb0ae79f79f93c : SUCCESS in 1h 34m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5ab16bdf71a4375b0a6a91ad8d64610 : SUCCESS in 50m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d9a372006a44225afa8c99b6011818d : TIMED_OUT in 41m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ab5007e3ddf04e18811b5b8ff820f3a7 : SUCCESS in 4m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7e2066d172f2474bb7a864b35346af1a : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/156d72fe474d4ad88eb874db3918ddb3 : POST_FAILURE in 5m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e9db31e45b10495b8d4c3fa28abe13cd : SUCCESS in 5m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/72341749811f4ee69522a3e00680fc7d : FAILURE in 20m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b371c6d632b742609ebac3f9c448961f : FAILURE in 4m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b4bf8caa3c0948e5b816d33f1e85e3f4 : FAILURE in 1h 40m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/29188bed891a4e278eede86169f25cad : FAILURE in 1h 26m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b1b31c0abe0f47f49a4c26c3eebc881c : SUCCESS in 1h 57m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/55b1ae04179b4573a5a60c57f16cb690 : SUCCESS in 1h 49m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/30a086c3cb424f4e91824095e446e9f2 : SUCCESS in 38m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0dc3a577c12445f9a55d84976c3b3481 : POST_FAILURE in 1h 08m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3102425df4724cdbb2c3ee73a529adee : SUCCESS in 1h 41m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3ca6ddf415df4d28ba9397c9f38975a1 : SUCCESS in 47m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2f3cf797d16f41899a1150bd0731bdcf : SUCCESS in 2h 24m 30s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/21f8b65b73cf4fb78f606e4527615905 : SUCCESS in 1h 14m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c08726e3feb045c4bfd7da981590d13b : SUCCESS in 53m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/411e6ae8684a4c418e1fd2f50c6025d3 : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":48},{"id":"347ca9245df6da9047394f3fd248ce64b8676949","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-02-20 13:26:23.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"2ddd420ae53a23cf18dfe9d4dd865684f17fdc11","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-20 13:27:27.000000000","message":"Patch Set 48:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 951157,93","accounts_in_message":[],"_revision_number":48},{"id":"db0fb04f8d75984746c4246bcfac4c29e58b5aac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 13:46:06.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98d019fef6f645e486e9e7aab1bbc80f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5248281e5a0e4d3dbdd60d88dbdc191d : POST_FAILURE in 14m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/43324eb605834026bdad8129f62b2a45 : FAILURE in 13m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9bd4582af6424648ad9299a90489422b : RETRY_LIMIT in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"51c767b6a84c8d2c1591ae2283aa0a177550d1c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 16:18:17.000000000","message":"Patch Set 48:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75708b50212e461d821b44c59ab6b3f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfe3e3b56c3d4181ace3821fedc7cf13 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b6e5d2b13b14748a0ae7edd5a0a43ab : FAILURE in 9m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f85d67254ad47f6bbe5fb05629ec65f : FAILURE in 8m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9efdc225b208453bad98fc01d62d2299 : FAILURE in 14m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed6d587d235747cabf9edcf4aa411cc1 : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/c8f5eb22416a4ceaa7c93f3037341045 : SUCCESS in 1h 07m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b33e07d9b5ba47dcb4e26f51dfa44401 : SUCCESS in 1h 03m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a12311b516a453a893d5347e03d9689 : FAILURE in 1h 22m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fe7455db8974e019ad0e7aa808252ab : SUCCESS in 1h 01m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b9be4ee6d02247d79ad5a828feb7cdc8 : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2279f96b7b7041c389cf3e3143295f43 : SUCCESS in 3m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/28bea35ee92748e0abe065790e2b633b : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a3daf8a61672496e9cae9a96767933dd : SUCCESS in 3m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/70b3584ac19f47ae8dea4b1febbc8c3c : SUCCESS in 10m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/29b63131d7924d40a4da53dcd99741c5 : FAILURE in 40m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a2a9816a4a7e48988ede2c60c3255bfc : FAILURE in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/961e891b9f9e4a4f9996c1b2446629fb : SUCCESS in 1h 48m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f8774498e8714621b7b2705da378d592 : SUCCESS in 2h 17m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8c89aaaef7e340c8b0d2a7f1148a39a7 : SUCCESS in 2h 45m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b591bb54378c4a6086131d19a780947d : SUCCESS in 1h 52m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/54ce69b541a6419eb2bf328f14653de3 : SUCCESS in 35m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b3b631d8608c49d3ad6bccb3105c32df : SUCCESS in 40m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6e20909e888e43398c050518d3e68236 : SUCCESS in 2h 08m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c50170c84f2f4ac3ae18fd531d8aea4d : FAILURE in 44m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/83e1c2f4b7344597ac8f22ee20a122ab : SUCCESS in 1h 51m 47s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/049297a53ee549faa750a4acf2f8c65c : SUCCESS in 2h 15m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1bb0e47ccac3402388aeb8f066268161 : SUCCESS in 1h 39m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b29e8f6241404605b43a3208bb79d12f : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":48},{"id":"4f26bd05432ad46bf4ee7128adf1185b7825af83","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-25 14:43:23.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"4751f171e6ff189cbd2abb30127bb91103bc63bb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-25 14:44:27.000000000","message":"Patch Set 48:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 949006,81","accounts_in_message":[],"_revision_number":48},{"id":"c7483a7fea4d9abaf23a56be9710dd31bb30e40d","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-03-06 18:06:34.000000000","message":"Patch Set 48: Code-Review+1","accounts_in_message":[],"_revision_number":48},{"id":"18f704a61123f56f4f83fdd124331e9ec1e2541f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-03-12 05:55:33.000000000","message":"Uploaded patch set 49: Patch Set 48 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":49},{"id":"efd3c9863f941f00c6dbf6e0639b6fbb8d341f87","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-12 06:01:32.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c855cdab7b144661a0f8183d23a529e9 : RETRY_LIMIT in 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/18419f74c5794fb79e51e078847ce555 : RETRY_LIMIT in 35s","accounts_in_message":[],"_revision_number":49},{"id":"3867acef3927a63366badd32ab4f8b1162d1f6dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 06:16:13.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8fac07f213d4fbbaccc8dc9648fd641\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9c36e31b3c64cb6b0bda8aa5af154b5 : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8b8604905d9c469f99f67ac3e1024a38 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1590e956418c44e088d9fcfcabc40669 : SUCCESS in 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"df304aa0fd68ca8482f9da9c15fe65854828fc58","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-12 06:34:55.000000000","message":"Patch Set 49:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":49},{"id":"e6c23b0c8af6165c9b995b1a4ba1fe0df1b61fd4","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-12 06:45:50.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/c59bba36117c4d838036618c5ae8ca60 : FAILURE in 12m 19s","accounts_in_message":[],"_revision_number":49},{"id":"3a868b084b25d6fc358d6b5f766e121e3b4ee781","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-12 07:46:01.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- hpe-fc-driver build/fec50c66d09b4a33a22780290a0c463f : FAILURE in 31m 49s","accounts_in_message":[],"_revision_number":49},{"id":"0fd6ace12891366965fdda1acea6665b95095dd5","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-12 08:13:44.000000000","message":"Patch Set 49:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/958465 : SUCCESS","accounts_in_message":[],"_revision_number":49},{"id":"7837bff292bb2a73025ae7f8151b7ff0b9db0846","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-12 08:17:31.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 36m 54s","accounts_in_message":[],"_revision_number":49},{"id":"cc76c744e25b1b5747c80856c5b267cd0276cb82","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-12 08:19:49.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c60c89967a46454a9d70a818f50b7b1f : FAILURE in 37m 44s","accounts_in_message":[],"_revision_number":49},{"id":"7c895b1d53be30ca278f9a0879b471df671d4348","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 09:12:00.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7f081c5d4b149cdbe040231176497c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/607631c5d4474b13b83126df2c654c3d : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/900e60f2b54f4d7c80fc669f8cb906fa : SUCCESS in 11m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cb5b92a4ca84d8f8f20f60404fe6821 : SUCCESS in 8m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9ecdd600095a443596beea1bda871253 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f46037656148f8b913c39fe9904153 : SUCCESS in 14m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/bf994035d76c4a8e91153be1572b8469 : SUCCESS in 43m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ba219bc10b9547b8817ca795da59a2fe : SUCCESS in 50m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ae41e31e5c7e4c159b2950fd45362d19 : SUCCESS in 1h 58m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1644b761f0404b08a764727fbf0afa3d : SUCCESS in 58m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/03d2dc19e9b849f49b15ad22ad8a32f6 : TIMED_OUT in 41m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7d1197db18a428f9fb8c0ce3bd95e7c : SUCCESS in 7m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/09248bdd311349769fb636e745129b14 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/870a0a5f060946c9844f1c947545c32d : SUCCESS in 7m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/438899cc6184489a84d81bb77f494ad8 : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/666fc91e653f40dfbcd3c05eae802b21 : FAILURE in 40m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a4e284a99da4c8aa4d4fea641665584 : FAILURE in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/74bcf8ace50648ffb2ed5d7db6924bf2 : SUCCESS in 1h 47m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0e1fd860f3ba47f0afd0328ea7dba694 : SUCCESS in 2h 18m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7e72cb0eccca4189ac47a40f7d531a7d : SUCCESS in 1h 59m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2bad97e5ad4d484883b1982ca0bb6ccb : SUCCESS in 2h 05m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8230973c3a4548d9b3ce66f1a7c5b2ef : SUCCESS in 30m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/da67f4550033447b8a4cd9f2a6a8f995 : SUCCESS in 37m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7625330445f1450e9185843e8568fdef : SUCCESS in 1h 32m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0e788abcc052458ab9e1989e292d5892 : SUCCESS in 45m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a50f9fe6dfbd4fecafa22def42a19e24 : SUCCESS in 2h 08m 53s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/8a1ae930af7845599e77a779825b38f8 : TIMED_OUT in 3h 09m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4bdccdd98e2646439a18f5006fc5295d : SUCCESS in 1h 20m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f4fd61fe8ad4d0f8c0eb7262e241ede : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":49},{"id":"d3ae7f7b3685c471f296607bea466391ff89bab2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-12 10:23:54.000000000","message":"Patch Set 49:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3453/.","accounts_in_message":[],"_revision_number":49},{"id":"dbd6d3182466d262c340497d020afebc92b79f99","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-12 21:40:25.000000000","message":"Patch Set 49:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1082/index.html : SUCCESS in 93m","accounts_in_message":[],"_revision_number":49},{"id":"d3e9ffe03688f711b48efde2810739997ee3cfe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-03-17 10:55:03.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":50},{"id":"8bf124fa197b862e9cab850f3e78edd32f87d5d3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-17 11:15:15.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/18236d6d7e564e01932048f21b8c3c95 : RETRY_LIMIT in 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e6e6149f13b1418a89cf066432034916 : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":50},{"id":"0aba08f16998eab59ab573d3f764470a656c6fd8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-17 11:35:25.000000000","message":"Patch Set 50:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":50},{"id":"c182b6c459f04d2f58ccb5a0c6b107165a525b4b","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-17 12:01:41.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/2f6c7e44732e4ea09fd3eacc71e6df11 : FAILURE in 12m 46s","accounts_in_message":[],"_revision_number":50},{"id":"cb7bf01e0c44e3716224d69ebfb0c001e038f8a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 12:47:55.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64c491bbd3164167abbdfdd3d4086761\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fefec41cc2784f6caab41ce8d8ab91f4 : SUCCESS in 17m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/621b580e8a624bc3b4005d62011e9cc0 : SUCCESS in 23m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/31b896ec53854cdd8c2547af9f7745eb : SUCCESS in 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"a73ef3b22fddfe9a7281b71b097366a9bc3ef5ef","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-17 13:11:18.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- hpe-fc-driver build/fc079fa317a94e3783d98b7f0e922b9a : FAILURE in 33m 40s","accounts_in_message":[],"_revision_number":50},{"id":"09ea7577e79f5d49c547df4cd3e4b766759c81fa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-17 14:09:05.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 39m 35s","accounts_in_message":[],"_revision_number":50},{"id":"c569c1dceb5e682b3459591cd290a43a50621c71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 14:17:34.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46b1d049d41142cfafe2d996b1fd84a7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e9375cad455435c87b5765b43e2a00b : SUCCESS in 8m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd1d275f008d45bcb198c9739bb4758d : SUCCESS in 11m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d925349a80994f719f683719a5d74734 : SUCCESS in 12m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/80249d6fa27b452f9c803a3bebfd41e5 : SUCCESS in 17m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1097c714ea74e62a9757711358ef247 : SUCCESS in 15m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/25adc34b3718469e90025c3562f4b370 : SUCCESS in 54m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ae2042736db4a63bc43f2cd6c994527 : SUCCESS in 1h 05m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e95de3ad5252480698cd37ffc42e64d4 : SUCCESS in 1h 41m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02c62e0d0de1407cbf8b329dc916f03c : SUCCESS in 55m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6dbfa8565c324137b999e2ada15af19f : TIMED_OUT in 41m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aa3c2acf1cfc45198285b9dfa732cd5e : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e3389e15d1244cf1a52318634d5f2464 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/434f565a558e42b894ce98151fdc0b41 : SUCCESS in 6m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b75d7acad52c4427bbc2cd8f6aefd1b8 : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7e4552e88d1e4dd4947776042a4ee51b : FAILURE in 31m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f1f4818031b945e5b518d01188a1c4b7 : FAILURE in 5m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/248b30cf663b495783fde7ccb6febc81 : SUCCESS in 1h 42m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6aad68a5c13d42e1b8ff8843690486fb : FAILURE in 2h 19m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b5ab799388aa4402ac3d40d1b91403eb : SUCCESS in 1h 35m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c7fff5cb45b64590aa787500e6f3afcb : SUCCESS in 2h 13m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bbef6f78cdca4ad1b1a5c03be13e0404 : SUCCESS in 36m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6aa6f1085d014910b0068cd968b72f65 : SUCCESS in 52m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a7eee63a36df4f6f9cd7ff93fa966923 : SUCCESS in 1h 30m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/88875adf5c014549b3cfe56b7ae111d3 : SUCCESS in 41m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2cd6e4a830574d6ca09ffdfe482b2ca7 : TIMED_OUT in 3h 11m 19s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/71b45ae966324331bd69c66f6b56e2f2 : SUCCESS in 2h 10m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c638ad30615a46679a51f8b6aaa4b24c : SUCCESS in 1h 12m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f139567ef18840539246ed0844597197 : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":50},{"id":"2f9dd7c0751e7b5b0f6bc98a4f844814dceb116c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-17 14:33:30.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1a7d4b2017ef436eb7b7cf880a171b4d : FAILURE in 39m 41s","accounts_in_message":[],"_revision_number":50},{"id":"37529ee8f179217fbd107769e4bcf891828c0520","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-17 16:41:42.000000000","message":"Patch Set 50:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/958465 : SUCCESS","accounts_in_message":[],"_revision_number":50},{"id":"e354aa33d33837ddc2cd42d2b2a65423134e086e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 18:03:12.000000000","message":"Patch Set 50:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3513/.","accounts_in_message":[],"_revision_number":50},{"id":"3ee4b96a8a9b52051c03f194eb915f8fa868becc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 20:57:09.000000000","message":"Patch Set 50:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3340/.","accounts_in_message":[],"_revision_number":50},{"id":"683d68c40da8ffdbc7a3212ff437a5ded2b91d71","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-03-18 14:08:50.000000000","message":"Patch Set 50:\n\n(1 comment)","accounts_in_message":[],"_revision_number":50},{"id":"c97f82604db97652678626d64ac0af39d8aad232","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-18 14:18:15.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/377edde656c34b42bb0fbe82612a4c63 : RETRY_LIMIT in 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fc778165dc5949698d422db501535c0c : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":50},{"id":"e38e616d1dcdad0f7ce5ec7d08979a403ef210f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:37:01.000000000","message":"Patch Set 50:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/522e7a1c93e44661ad0a6955416806de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afcbd6f6e3fc47ea9786ea4955ddb6c7 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/712ed6cca212486eb563c01ac7c159f4 : SUCCESS in 11m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a49a4a7be8ba46609352c45afe82de39 : TIMED_OUT in 41m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/351dc6d723d54cb197ee9250471b13dd : SUCCESS in 15m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e29b80cfced0446e8e063f508cca22f0 : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/3b6b9a6d744d4e38a7fc7dd89a479c6f : SUCCESS in 54m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/623317c4398f4a0e955ab1cd463690eb : SUCCESS in 39m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c37cee6ff0444ec28f2dc59fa59a1237 : SUCCESS in 1h 44m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/633d7795cf1f40069efbc7bc89c70ceb : SUCCESS in 1h 04m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f992b75ed25d458ebd22d5af24c00965 : SUCCESS in 22m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2c44973a95d547fd89fadb92c9d38334 : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a4e46bf689c456f96ed09d74d603fa8 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45d53e8b19d647dcb33deb7d646d95f7 : SUCCESS in 6m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6b0470717b0545a8b36d9f830f444b65 : SUCCESS in 8m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/66e9e940a1fd4d57b0de98764ebd10ea : FAILURE in 40m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d6268dde00e34da295cf8338b8167c29 : FAILURE in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5233aca2777843f08be048937b442e67 : FAILURE in 23m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/96ac77954b074dd89aff1bcd9f9175cf : FAILURE in 2h 18m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e524b9711914d43a545772ce0d7a5dd : SUCCESS in 1h 48m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bf3a42e3fdae485483775546f2e768ff : SUCCESS in 2h 10m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/29f6f81fa69a4f3ab456f85a71b40b77 : SUCCESS in 37m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c8833e2c14ab4f23b35905efb4e43a17 : SUCCESS in 1h 08m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/85ed9d4f94e046dda2d2db441214024e : SUCCESS in 1h 29m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/260d26c267584543b24c9ebbba6d7235 : SUCCESS in 45m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ef9fabaf110a4577ac13f3eca135fb81 : SUCCESS in 2h 13m 50s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/05278717f2a14fedb3402aa76b191199 : FAILURE in 1h 23m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b43baea76d84871994ad0d901a8dfe2 : SUCCESS in 51m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b361851ab75842b1b6fbfb483287f2ad : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":50},{"id":"8cc75cf2b3180a24c9efbe593f405117b192cd2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:43:30.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2956db54d52c49e6a51a075399d06c7d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb18922cc9bc412494f5d33beb508be0 : SUCCESS in 13m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3f5022730856401e9659f59b42ffddfd : SUCCESS in 14m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/effb89cb59474e8fb7bc1127ee9483e2 : SUCCESS in 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"dac2faa40a789c58939d25c4395625b6b9300d52","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-03-23 04:52:41.000000000","message":"Patch Set 50:\n\n(1 comment)","accounts_in_message":[],"_revision_number":50},{"id":"717b9ba425adecdfeb8f695edf8513769fabb765","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-23 04:59:24.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/470ae0e513654d579479e5d3bdf52cb0 : RETRY_LIMIT in 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c993b97f47274a04b48857dd5313cda6 : RETRY_LIMIT in 35s","accounts_in_message":[],"_revision_number":50},{"id":"10a17e4093de7f64f560615ca906c6d5df6169bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 05:12:17.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80dd074b6d1a42cdb915285d53ccc0f0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f2883abbdcba457e9c6a742b30b914a4 : SUCCESS in 13m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aa24e326d519402ab582a04ed1a9da12 : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/796d280420bf4ff1aefa13bd6307c7c1 : SUCCESS in 15m 55s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"c03e253483039921e8af6b66a8359a2c7bfe2f5d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-03-23 07:31:17.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-50-fc : FAILURE in 52m 21s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-958465-50-iscsi : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":50},{"id":"ef62c6ea3f918a353f1dd6da4558f3e8425cd100","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 07:59:15.000000000","message":"Patch Set 50:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30ac546c7a1845cfbecd40fddd452b71\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b4bbfc94dd41efbd1d3ed60d09eac8 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a2cd2d84336435ebd7cb31a18ac2851 : SUCCESS in 9m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e33a7c036d44e2ba9ade09bdea50fa9 : SUCCESS in 9m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/43fc788af5e842f3be18c0d16fdbb9fe : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a77e154a1ab9439cb6aa267f99e20a6f : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/da986ddf684d4108a0111c9ebc762418 : SUCCESS in 44m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/280b8dd93999475d932ed6d222e5f81a : SUCCESS in 34m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/233820f6be6448efb714397a89e18449 : SUCCESS in 1h 24m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea09859deb1a45abbf4d5517c615185f : SUCCESS in 44m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2030599bf9d349a39a7601760008bc37 : SUCCESS in 20m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b72a1b7736c24c6482a1910daff7264a : SUCCESS in 6m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5ca4dc27bb23447a877494f4d5202b2a : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3284d94e93d74ae4a4c2e4ecf0cbb05f : SUCCESS in 7m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5b06a88f00e44f4c8e47d47b9588aab2 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/296c5287096e4386b82ac6e103c857c6 : FAILURE in 39m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e76fc46657444f9ca57f65ff1a036b43 : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1c98e29e988a400ca8c29af8aad915b2 : SUCCESS in 1h 11m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5fc6b2b8177b4ddba302220a4e415974 : FAILURE in 2h 36m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/85b18568f48949d6ae544200f7fb1912 : SUCCESS in 1h 54m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6038a781375349b6bdc5015d645f0101 : SUCCESS in 1h 49m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d1e25bd06887496fb71c8e4ce1e50dc2 : SUCCESS in 36m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/02e66d59173b4c2497a7a8a1bd56929a : SUCCESS in 34m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8c6f7937aab84810a2fd767b3c6801f1 : SUCCESS in 1h 28m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/503bf514e1c64162996ebe5fe981ea87 : SUCCESS in 28m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9a2ed2ad36354675b99f5028e4c0e38b : POST_FAILURE in 2h 58m 45s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/97961f70bba9460aabb9198b05bdaef8 : FAILURE in 2h 05m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f86753feabf6412e86eb66fdc18b7b95 : SUCCESS in 1h 40m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69b28a7bb6fc49b888daad1265e48ea3 : SUCCESS in 40m 21s","accounts_in_message":[],"_revision_number":50},{"id":"e5e658112f904746f92c9eff9fd71554c0a49e4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-03-31 05:34:39.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":51},{"id":"214c9bb13d04bad692cdcc63cf66b5d616572532","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-31 05:40:30.000000000","message":"Patch Set 51:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/6b53f50b0e9b46248e5ab286fe3cb737 : NODE_FAILURE Node request 200-0000692527 failed in 0s","accounts_in_message":[],"_revision_number":51},{"id":"fa8205a06fe357aa7c53f906f2f8940d9e7fafb6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-31 05:55:50.000000000","message":"Patch Set 51:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5ad5f394d5a14a7db3cd713c2688763f : RETRY_LIMIT in 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/deb3cfa688c2413ab6594327967888a6 : RETRY_LIMIT in 34s","accounts_in_message":[],"_revision_number":51},{"id":"3bf1bd4f25b3cad226c316d6d61186690fc68ef4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-31 06:14:24.000000000","message":"Patch Set 51:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":51},{"id":"914ae13d4a7d8c471461c28226ac4f1048703fbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 06:20:56.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38259495041a4e1b88d73ea575d6d39a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/786b5c2035ac4f41a374cc2501ca533c : FAILURE in 6m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ffc455d925f043bab8f3b10da539b886 : FAILURE in 7m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/00bda39b6c794687be2818fbdec9cdac : FAILURE in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"629e1ad1f922446ba89d8c7e29577e56b2b86c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 07:47:40.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b57697e1e2e4b458b29cfd7ab9a90ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a23f40ef19845b4a09bda792951fcc0 : FAILURE in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4197bbbd798344629f77079b9c17b3f5 : FAILURE in 6m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3fccae0c0889421dba220c2ce19e8e6c : FAILURE in 6m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1f214f7907c8459c9a1d6009de83e6fc : FAILURE in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2199913fef1446e587e444762aa9f98f : FAILURE in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/dbd5f9f1cf954adbbc4e355a3b6a438e : SUCCESS in 1h 01m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74fa36af18a54c9280d0b7299d063f61 : SUCCESS in 58m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17a64978e8bf4e58a9ecc8bd74fcf058 : SUCCESS in 1h 18m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b13dcdcbc5d34e938aaf0d7df8d1cc7f : SUCCESS in 1h 02m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/beceaab7d8b54d16931fa308976cf60f : FAILURE in 6m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5995d5a71f704e2bbdda83c1784b9e3c : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2a113abef04849c2ba03d06e8bf97819 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/24b041f0f5d84ce9b06c691e50b4e370 : SUCCESS in 6m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f050af17a86e4412860253fa9e57027f : SUCCESS in 7m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/176e10900667490d8866fc76039f0d33 : FAILURE in 24m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5271565ce4dc4502a142ccd164b7b02b : FAILURE in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/242c03d019ab4a8689b11d06c11edb65 : SUCCESS in 1h 50m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/61ae4b282be745ac985d69d321acbed2 : FAILURE in 1h 29m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6da3aacc24064e9f90727719e1a18666 : SUCCESS in 1h 47m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1442b28bc76b4ba59d1b401054f1168c : SUCCESS in 1h 59m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/526ebbb7f7694e7293b9cb23bad66262 : SUCCESS in 24m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f8fafd54bce54f6cbe656f43012e83e1 : SUCCESS in 1h 19m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/37b67d9d5a45444f9b1fd948bc7edd97 : SUCCESS in 1h 34m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/20ef36f5590a449a823610fe46caa6cd : SUCCESS in 32m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b9696abd0ee944a8a4d7b3c14ef507c5 : SUCCESS in 1h 28m 37s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/5a6df7f7137c40159fe355076aa41fca : SUCCESS in 2h 04m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa9912fc870d4c278ff2b2bdef87c15f : SUCCESS in 1h 53m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/136bda14b24b4f6cb9bb95ec241f43d2 : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":51},{"id":"0cfc393a2cef9a581dbc42523842ff58ade29e16","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-31 07:55:31.000000000","message":"Patch Set 51:\n\nBuild failed.\n\n- hpe-fc-driver build/ab5c19a9f11d41b2be233cc9ca42c9b1 : FAILURE in 34m 30s","accounts_in_message":[],"_revision_number":51},{"id":"6cbcec4418f03aa596afcb159f9573bafad0dfde","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-31 08:16:39.000000000","message":"Patch Set 51:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 39m 09s","accounts_in_message":[],"_revision_number":51},{"id":"8936ea16c701ce3c1a6b260ca93b6a9d77d09454","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-31 08:27:05.000000000","message":"Patch Set 51:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/859aff8b5d5b4b748a9dac9c8806e0f2 : FAILURE in 42m 38s","accounts_in_message":[],"_revision_number":51},{"id":"c950a3e8497b32109739b64b1d20f44bc1283754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"real_author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-03-31 09:40:01.000000000","message":"Patch Set 52: Patch Set 51 was rebased on behalf of \u003cGERRIT_ACCOUNT_36800\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"}],"_revision_number":52},{"id":"ab2cde51490ef37daebe5aeb9a0abad4f3967d58","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-31 09:42:39.000000000","message":"Patch Set 52:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/1288cad4f73b419f9c9cfe0881485e87 : NODE_FAILURE Node request 200-0000693067 failed in 0s","accounts_in_message":[],"_revision_number":52},{"id":"1e88b1f4e1ebc86932014f1af7e2685aafe0db6b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-31 10:06:09.000000000","message":"Patch Set 52:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2f67c764520b4340a570f92a34430c4e : RETRY_LIMIT in 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1dbde27e3fd14a458ae46a422ce209db : RETRY_LIMIT in 37s","accounts_in_message":[],"_revision_number":52},{"id":"7677e8b8309ea7ce11c48af9684e8482f98e2c0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:07:14.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be97ff11e0a042a69d8931b53faae631\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b36a96da48f2414daa8af6301536aa2f : SUCCESS in 13m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/58eb8362aede4751832981dc7c06990f : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c1d8e7d360484fd2856c21e2a23625e6 : SUCCESS in 18m 42s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"0dddf21278b75f23208b5e4a5906ab6d66ac41a7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-31 10:18:30.000000000","message":"Patch Set 52:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":52},{"id":"cffb7818c482755e1890530b5732175a98a9ee8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 12:50:18.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c9702bc28540ebb248c49046eb6bd1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae2bacbbf39b4786a2475b651bcfd6e3 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba52a7b656434733ad52f22f2e352016 : SUCCESS in 11m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c6bdec3e8b04a858b65447c16d94e75 : SUCCESS in 11m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6b38e5a06555437bbde96c3d667f4cce : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5377a9c77dd4f07ae147773dbe31290 : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/78ca763487c44ec2993a4c46cfd668e3 : SUCCESS in 1h 01m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f464c56406248859791b529f2ac07e6 : SUCCESS in 1h 13m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/03323d0c26594534b030da0e810df6a5 : SUCCESS in 1h 38m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e9c5d5cdc4e4edb838d161872a57d84 : SUCCESS in 1h 05m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/01c004335c5d4b659dc59cef0fd1e36c : SUCCESS in 26m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d9b9af7181544dcea377e6ec85415a34 : SUCCESS in 3m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f81022239b54695b470da869efebddc : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d38eac00b90f4aa69670cc95eb0d52bb : SUCCESS in 6m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cbaa81bbae63434c997c7a90d2a66a90 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6ddb9979997c4ebd83f86f2c4441365d : FAILURE in 30m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e641fe5c9134ede8bdd4ded7c4c8d98 : SUCCESS in 4m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a13cfa3a875641fcbe0f0fb39b6382a9 : SUCCESS in 1h 02m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d5af5b912abd474ea2a0be118de5c840 : FAILURE in 3h 01m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/63295cc7fe674d20b7437d90036e03cd : SUCCESS in 2h 15m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2a263456009d45f0816341966cbd4b44 : SUCCESS in 1h 46m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8fe704a95b6846a4a9b2540816e2ee10 : SUCCESS in 30m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5c33998a6fa64233933057f117bb6f30 : SUCCESS in 1h 20m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/888c14be52c74e03ae57c74296398874 : SUCCESS in 1h 38m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aab4951a3bdf4e21bcbe2b07ac2a266f : SUCCESS in 46m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c2698284b155456bbfd4e3fce4ecff77 : SUCCESS in 2h 34m 09s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/612c45124cea4a2991a115ec553bac39 : SUCCESS in 2h 05m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/083f1e25eaf14ebd9a892c61ab1a53c9 : SUCCESS in 1h 43m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe4c6e0abb3443ff9b4a814d4ead5043 : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":52},{"id":"3b786852f9d664d6632400182e89b11425fcfd9e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-31 12:54:08.000000000","message":"Patch Set 52:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-958465-52/FC/ : SUCCESS in 1h 57m 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-958465-52/iSCSI/ : SUCCESS in 3h 8m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":52},{"id":"d5dc68abefb8a6717e69dba5aead3dea00cf9697","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-31 12:55:59.000000000","message":"Patch Set 52:\n\nBuild failed.\n\n- hpe-fc-driver build/d039996a0d5a4c568590da8572c01c92 : FAILURE in 35m 11s","accounts_in_message":[],"_revision_number":52},{"id":"4e0cf2fef7d30e37a9f1b49bbaafc9e44f55a0b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-03-31 13:00:08.000000000","message":"Patch Set 52:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_03_2026_12_53_36\nLeave a comment with","accounts_in_message":[],"_revision_number":52},{"id":"e5f24477b053ec2c7bd327d7ad3d5266abdab664","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-31 13:48:06.000000000","message":"Patch Set 52:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 38m 22s","accounts_in_message":[],"_revision_number":52},{"id":"a7f24b390eb4b99fa8d645e8879a576f6b2ad5bc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-31 14:07:38.000000000","message":"Patch Set 52:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/d0e2e057a9db432c934492047c8a860c : FAILURE in 40m 52s","accounts_in_message":[],"_revision_number":52},{"id":"ae7e27038db020d3effd3c976da465b95c513038","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-31 16:32:09.000000000","message":"Patch Set 51:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1289/index.html : SUCCESS in 103m","accounts_in_message":[],"_revision_number":51},{"id":"8ef4f4d385dfab4fd8431f315e261181233e5ba6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-31 21:27:31.000000000","message":"Patch Set 51:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3638/.","accounts_in_message":[],"_revision_number":51},{"id":"78fe83d04ea876a0d809693d494408ed38d6e10f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-01 09:53:08.000000000","message":"Patch Set 52:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3648/.","accounts_in_message":[],"_revision_number":52},{"id":"2cac34d90f76b9cff6d8711b9a13d5e6749d87cf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-03 15:57:28.000000000","message":"Patch Set 51:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3465/.","accounts_in_message":[],"_revision_number":51},{"id":"3e5854268863e1578290607532517f077df24429","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-04 01:52:03.000000000","message":"Patch Set 52:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3475/.","accounts_in_message":[],"_revision_number":52},{"id":"6e9d0a844fdd0532aef942cfe4b2c7afe0321b22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-04-09 10:19:31.000000000","message":"Uploaded patch set 53: Patch Set 52 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":53},{"id":"163a196d68e4517a5ce8e74c00d7d5f6ba4fd74f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-09 10:55:59.000000000","message":"Patch Set 53:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bd5926ca07504c40aa66d71d965226c7 : RETRY_LIMIT in 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f613752c97354e958aa1a40d70a38d5e : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":53},{"id":"387ea5c7e12019f85caea300fbc3c31982773003","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-09 11:02:08.000000000","message":"Patch Set 53:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":53},{"id":"e41b2f64afd5146ebb7fcffffd199db1ee47f473","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:40:13.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1c704154e304a80b28d4f7ddbb4d615\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e33ee9b548904a8f8989fa5baad35c66 : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a48e1fc391b44ce8989fdfb415813e33 : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/886dc1e8446f4af89988ae31c148d7e6 : SUCCESS in 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"96525b3622fe6b7e9bc2336c2eed49b6ebff3238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 13:58:26.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0525940952a646d094c0c5ed3e4da570\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/280d9dc725824d17a906bf39f6bca371 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9174fd05aae46e9acd110674357fe19 : SUCCESS in 10m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/58f27e375817426eadffbb4ffb1020e1 : SUCCESS in 10m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96775a4d522e47e5b212a21e27be2037 : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58386f386d24683acffb2868ae1907b : SUCCESS in 13m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/9366fbb312f1412ea2a8ed894c06ad6e : SUCCESS in 33m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94df2720285e4d0ba78b10e5503634f7 : SUCCESS in 1h 01m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f9d11ae84e724b8d9f3b183859c7d42d : SUCCESS in 1h 31m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ee17be47c134b8f90bd9c0e09ed77a3 : SUCCESS in 43m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/51f045a145504d60ac29a40828949c5f : TIMED_OUT in 41m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7cf6085522e1452fa6ef73099c22a629 : SUCCESS in 4m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ef8e9cc6eb8b469b8a79f21397a624e0 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7411ab86ed664f0893f69d24b3c0a264 : SUCCESS in 6m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/356d64544a45430eafeb288c1034a32e : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2436cc2a39144ab98b56bf168c92b346 : FAILURE in 33m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b059730d5bec4b63a4c6d6579c4b61ee : SUCCESS in 6m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/adff340c0b7945e8871a1d315f18d556 : SUCCESS in 1h 48m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/840574deddd34590a32694c78b8b5f57 : FAILURE in 3h 03m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f90ba694a17411f9abd9ae3e678a2e5 : SUCCESS in 1h 58m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5c874f627d6446ef8f22d836e1a838da : SUCCESS in 2h 17m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/95ba1da6c3794a5eb7cdc507c98979f2 : SUCCESS in 43m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1e71f4e4f89b48feab1c0147af891596 : SUCCESS in 1h 22m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/09a7b8afe1b246e3a3dab44584b6d69a : SUCCESS in 1h 45m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/51b5cf24a5bf4261b4be4798c522e4f5 : SUCCESS in 43m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/95fa791be6db465c8b126fd118a8649d : SUCCESS in 2h 19m 59s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/85980a7e77a54cd9a9b36b1925adc6f0 : TIMED_OUT in 3h 35m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/11705aef99fb4868a275be8d53c33091 : SUCCESS in 1h 44m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21667f1a624a44b8857dee46102b7111 : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":53},{"id":"af98b0c7246f3175cb954e16fe6cdf98b6b3bd4c","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-09 14:00:03.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- hpe-fc-driver build/65f48863edd545c9a3dc617919c7aa57 : SUCCESS in 37m 43s","accounts_in_message":[],"_revision_number":53},{"id":"6f815b324a2bb883d3df5c027ca93a9a6ba63d47","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-09 14:08:09.000000000","message":"Patch Set 53:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_04_2026_14_01_40\nLeave a comment with","accounts_in_message":[],"_revision_number":53},{"id":"017904afbb54f436d501452370ee95c5148b48f6","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-09 14:23:44.000000000","message":"Patch Set 53:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/4614068627094ab4b24affe2b4389cc3 : FAILURE in 35m 17s","accounts_in_message":[],"_revision_number":53},{"id":"2647535537a6823012ae41aaa47d36768d9cce46","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-09 14:27:10.000000000","message":"Patch Set 53:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 38m 44s","accounts_in_message":[],"_revision_number":53},{"id":"dd6f0283426a7903ec4ec80e1ca727037023732c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-09 15:48:37.000000000","message":"Patch Set 53:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3559/.","accounts_in_message":[],"_revision_number":53},{"id":"87e0da89a7195f1da3871b7a1b618f3e8e5d6e5f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-09 15:52:14.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/f9f8c81dbc2a4cc9b6b861279cf1faf9 : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":53},{"id":"371b563476963e79841540de7f0c6f7e87bdc1f2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-09 17:36:51.000000000","message":"Patch Set 53:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3735/.","accounts_in_message":[],"_revision_number":53},{"id":"c02b10fad56581b5f4c830ab1c34d881a6a70c1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-04-13 07:27:00.000000000","message":"Uploaded patch set 54: New patch set was added with same tree, parent tree, and commit message as Patch Set 53.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":54},{"id":"466f715aa4cac03c98ce2ecbeab8ce5133e5f155","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-13 08:08:39.000000000","message":"Patch Set 54:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":54},{"id":"fc868ee161156e1e0965697651718aabea390298","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-13 08:25:31.000000000","message":"Patch Set 54:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/90910edd4ccc4dddb0f4dc5ee44bde48 : RETRY_LIMIT in 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7b782ac1a4c2456b8ccc3fb820c92eb3 : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":54},{"id":"b001638b02ce918a9422eb86617a06704ef11aa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 08:57:03.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6042e4270f44e71910f43e807f035de\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cac7bce197cc44378c1c19b9db4103fa : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/588e6f7c02b946029d6d95bcd5aeb57d : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6186909570354edca4e1fa889b2c0150 : SUCCESS in 19m 22s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"6a713d8857798ab64f12978d1a63d1be5cceb984","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-13 10:06:05.000000000","message":"Patch Set 54:\n\nBuild succeeded.\n\n- hpe-fc-driver build/c2bd0831dcf04b378d18c06a18de931b : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":54},{"id":"361439552cdd43bd87decb1ab34ab9190fd9b08a","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-13 10:08:29.000000000","message":"Patch Set 54:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/adc8e7b7365e4b0a986f8c680cb1ddd1 : SUCCESS in 48m 02s","accounts_in_message":[],"_revision_number":54},{"id":"73819af535dd65ea40180937c459f36fe14da8e4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-13 10:15:23.000000000","message":"Patch Set 54:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 13m 55s","accounts_in_message":[],"_revision_number":54},{"id":"af4471a60f9a8c926058910a0099a1309f5739c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-13 10:35:39.000000000","message":"Patch Set 54:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/78423a1b7b464a89b531dc7c1a14e99e : FAILURE in 14m 54s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/958465/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/958465/etc/tempest_log.log","accounts_in_message":[],"_revision_number":54},{"id":"889eef181aa42516f3618ecfc59b014e8c32c555","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-13 12:53:51.000000000","message":"Patch Set 54:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3764/.","accounts_in_message":[],"_revision_number":54},{"id":"29ada2d94a1e24f3b1777bb1d17ba81e4b32e70e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-13 13:04:13.000000000","message":"Patch Set 54:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3588/.","accounts_in_message":[],"_revision_number":54},{"id":"06c8ab7360f272b1eadabe7de4ab02ecebe84327","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 14:22:03.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0592081d50854e7db49847e6284e201b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba6466a4ede46c2b1722fb603c18d48 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97b01df270e14ec4822059f17e7ce829 : SUCCESS in 10m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be124d5f98b742af915c7013b3d1f191 : SUCCESS in 8m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b6a955626f9440db80e912133fe78c7 : SUCCESS in 13m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb4b036d09f449397324f55f88495a3 : SUCCESS in 14m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/e6755d4c11ac481b8cb72d5de04b82db : SUCCESS in 40m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd336c9711fd460b90b3b7827f8140d0 : SUCCESS in 1h 07m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bcd47f7c522f4ff69a8f7c2ba3f2b67f : SUCCESS in 1h 44m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88220b7cf8054094988d937f8c3a5804 : SUCCESS in 1h 07m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b68a55305b064830a704827b56cb7eda : SUCCESS in 25m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f3dfc39211d445469c8dc1981db8e7f8 : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e0260a3609474a5e9eecacf70fec32b1 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e9dc7aa4d5af481cb376c8f59c1f26a6 : SUCCESS in 6m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/60c00c629a3244d99611169fc2c051f0 : SUCCESS in 7m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0bfb93f783bc464184a06236353ea621 : FAILURE in 45m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/975c99bcb3074a8ebc6eb97d82069491 : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/577c572448d240a99861c142123f60ae : SUCCESS in 2h 00m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/80c89d2e0cfa42008d19c52ddcc4acfd : FAILURE in 2h 48m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2c0e20a09b744ff0bd031f19bf838ca8 : SUCCESS in 1h 57m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e7392515f6f3462ebd0d155f2dd612e3 : SUCCESS in 2h 12m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6e28ece7677b4388ad67afc658b9d389 : SUCCESS in 41m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1335957752a54936afc571e4262a3662 : SUCCESS in 1h 14m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b5e61ba46b264d94b97bae3fa4fdbd55 : SUCCESS in 1h 28m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a4d1cefcc2864550a1df976fff8e142b : SUCCESS in 46m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/26a1787b126b4e81a739bfdd33263969 : SUCCESS in 1h 50m 28s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f8128c566a8a4f6d816d1e571cd75c96 : SUCCESS in 1h 19m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7371d177fc1e46dfaecf241285902b41 : SUCCESS in 1h 40m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4c96cda8356496eac3f2e9c52162a5c : SUCCESS in 1h 09m 03s","accounts_in_message":[],"_revision_number":54},{"id":"0d2342662505f40b77dfe624558a735d96447c9d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-13 14:42:09.000000000","message":"Patch Set 54:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_04_2026_14_36_05\nLeave a comment with","accounts_in_message":[],"_revision_number":54},{"id":"21863600c2714e160f091d1fd0d03d9712b91f7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-04-14 11:04:51.000000000","message":"Uploaded patch set 55: New patch set was added with same tree, parent tree, and commit message as Patch Set 54.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":55},{"id":"2390522108ad59cb2754d5b0d31db0e5b8707c96","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-14 11:47:27.000000000","message":"Patch Set 55:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":55},{"id":"06b0601b142d1d151c741fb5f6b33ca0c75d638e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-14 11:58:52.000000000","message":"Patch Set 55:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c4d013f59a5b4bfa9d587893781e6afb : RETRY_LIMIT in 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8b91daa3407e49a3b17bf18951681b19 : RETRY_LIMIT in 35s","accounts_in_message":[],"_revision_number":55},{"id":"d1370a006f3b2d1a5df46776e99cdad446e37dc0","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-14 12:48:02.000000000","message":"Patch Set 55:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/4358208fc65f43c89225ad4c71904260 : FAILURE in 22m 37s","accounts_in_message":[],"_revision_number":55},{"id":"70a7fdcc096330e5de7937d7e1fbcd6fa679eb62","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-14 13:33:25.000000000","message":"Patch Set 55:\n\nBuild succeeded.\n\n- hpe-fc-driver build/67b329bac5ec41c6be32cbd4cb4ce2b3 : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":55},{"id":"15f84716144f213a55cffb963c2485b70959bd7b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-14 13:50:27.000000000","message":"Patch Set 55:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958465 : FAILURE in 38m 20s","accounts_in_message":[],"_revision_number":55},{"id":"967a979425ef3c561f88fc106d6aea32739a0095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 13:54:47.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a9604172c4d4d8cbcfa541489bc964a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb8d28adbf2e4e4f9ebee8946a08eb9c : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e467e6950d843ccbdc6af410fad206f : SUCCESS in 13m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77eeca4bb0cd461388589ca9bb2b0341 : SUCCESS in 12m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a2548fbc766340a8bf896782898467ff : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0466ce22fb2b4946aceb51fc2943af4c : SUCCESS in 14m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/119ada966da841a5abbc46b71ecf74f1 : SUCCESS in 1h 13m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8195811054f3407c9cf4f7e8d4489ca2 : SUCCESS in 59m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7e93894715ac43bd90c82616ab3971cb : SUCCESS in 1h 32m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2b5ba8f268a44bbb9e8e00b99801913 : SUCCESS in 1h 20m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6cf5821cb0d44a20b706a6d8c3d1c32a : SUCCESS in 23m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ac6b340efe6941de859cdc697b7f59b1 : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a109067a1af430aa977f46198c36f04 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3173369761d343539cf9ad5bee5d083d : SUCCESS in 6m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f85f101ff6af41fb8a45ac7daa7534c0 : SUCCESS in 7m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5b4814505c2247899f8128535fd035b8 : FAILURE in 44m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b12c987eed1945eeb7a45a202211bdf7 : SUCCESS in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/752f989427414072915f6cc6368add0e : SUCCESS in 2h 00m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/40cabc43be734755a079e0c2e4be136f : FAILURE in 2h 36m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f1891dcf0eea40a19b3dd638096bf604 : SUCCESS in 1h 42m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/90964960ecfa4fa081b49338ab5bc811 : SUCCESS in 2h 25m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/04c6d4f2f728411ebdbac3455846870d : SUCCESS in 39m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b3fe01fba36c4fdba88ac8dc037b0bb4 : SUCCESS in 1h 12m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e5d6cc81f0504451870079c4a4528a9c : SUCCESS in 1h 59m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9e0263eab83b40388292ef156bd0d93c : SUCCESS in 46m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bcf60b4a44714824a4f1bcdcfe94fa6c : SUCCESS in 2h 11m 26s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/aefc50d4eb054240b6508cb755473307 : SUCCESS in 2h 26m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/48fd7f3b9e3945dfa91419f5150455ea : SUCCESS in 1h 50m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/041f19e4e51d4d55a23f00f4c4d4720a : SUCCESS in 1h 11m 20s","accounts_in_message":[],"_revision_number":55},{"id":"ffe7f3dea660f9187d91735f85b103fca2e0ab23","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-14 14:04:06.000000000","message":"Patch Set 55:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3618/.","accounts_in_message":[],"_revision_number":55},{"id":"72be2224fed4f0fb6c35e14b4b848c535b0ab809","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-14 14:04:21.000000000","message":"Patch Set 55:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_04_2026_13_58_04\nLeave a comment with","accounts_in_message":[],"_revision_number":55},{"id":"bbe851a108cb38f5e2acef67c89be408a7e67c88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 14:04:21.000000000","message":"Patch Set 55:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d14adb1425bd45fe85f183419c02fed4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f5bc075ea1c4d7e8fb82b418e72b4e8 : SUCCESS in 13m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4917b7f45b474512863a9e3935cc3a25 : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1301fff2a2d146eb8498517d944bea00 : SUCCESS in 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"80e604d05b7217e078f7903f8b28148a9b7ef7ca","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-14 15:18:25.000000000","message":"Patch Set 55:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/ec616b8ed2034d23b445052411c6a18a : SUCCESS in 50m 09s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/958465/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/958465/etc/tempest_log.log","accounts_in_message":[],"_revision_number":55},{"id":"6bb4f169317e056568dcfb12a16179d3cd808499","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-04-28 17:53:21.000000000","message":"Uploaded patch set 56: Patch Set 55 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":56},{"id":"5d8b524c7b97ae10c485496774053c851fe263fa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-28 18:42:34.000000000","message":"Patch Set 56:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":56},{"id":"3f72e3e4d51ba0421f77ca37cf475e665eb43957","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-28 20:12:45.000000000","message":"Patch Set 56:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/748d1b0009954670a628b580e411faab : FAILURE in 37m 17s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/958465/etc/tempest_log.log","accounts_in_message":[],"_revision_number":56},{"id":"214307cfe5555f85f7dd1454c1f5d4fb0e575168","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-28 20:27:10.000000000","message":"Patch Set 56:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/958465 : FAILURE","accounts_in_message":[],"_revision_number":56},{"id":"9d6dd72780e1ec4041ceacd044dfc6e1c5125935","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-04-28 20:31:10.000000000","message":"Patch Set 56:\n\nBuild succeeded.\n\n- hpe-fc-driver build/8d7b29dc2fa54f4382a3c437697c6bc3 : SUCCESS in 38m 20s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/958465/etc/tempest_log.log","accounts_in_message":[],"_revision_number":56},{"id":"11ceb77c60fd7e2706f6fa7e2e4cf1feb52552dc","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-28 20:49:09.000000000","message":"Patch Set 56:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/958465 : SUCCESS","accounts_in_message":[],"_revision_number":56},{"id":"59b9026f19434a205ceccb546d2682d97530bb72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 20:56:13.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53b187111bd74627b8bd554337d9c947\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb0874c766d84af8a5290abdbe9cecfe : SUCCESS in 19m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/900a05f9fc0d4a3daa3deca60c2b16c3 : SUCCESS in 17m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9af88f947c3c4c4ab7833ba49b625bce : FAILURE in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"0aab6cee1c004715c119f8549864a00198d41800","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-28 21:39:41.000000000","message":"Patch Set 56:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3989/.","accounts_in_message":[],"_revision_number":56},{"id":"1d5e487b3656c50a054c3de8014a6ce91f63eae8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-28 22:02:46.000000000","message":"Patch Set 56:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3813/.","accounts_in_message":[],"_revision_number":56},{"id":"d3f703846a3a3ebdfa8d1479eccc5609f9a66e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 00:09:53.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c9a2870b3324eb8ac4d2712c3249aa3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/834fdf1701eb44fb8c93696160b801f7 : SUCCESS in 7m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cfcae9ec11d84225a35c33dd9d678567 : SUCCESS in 12m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94378b0b6f9b4d3eaceb1097ebcec2fa : SUCCESS in 13m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3238e82278ff42ce86df0f98c5bcc06c : FAILURE in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b922850bf1743389bbf238ed6b890a2 : SUCCESS in 15m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/da3c7bea7da44fbebb57c6d0948557c6 : SUCCESS in 1h 09m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/483eb692daee4ec191046dc91067fe54 : SUCCESS in 1h 06m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa65b9b4ef5843649162060feeca191d : SUCCESS in 1h 37m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4de09c6bf964015a2abc2561362844a : SUCCESS in 1h 09m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0eca17d4ae8248f1bb32bb399a0e4133 : TIMED_OUT in 41m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd1f7c63ab594baaa3c4d55f8b71d055 : SUCCESS in 5m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9ffba6b1636e4a3692abbc3913cd71bf : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2cc165e775a14851b7062dea103a2151 : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/870e10b35d714e56ae28e8af69f2350e : SUCCESS in 6m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/628190c26460472e82c585d8e77aef57 : SUCCESS in 34m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/97464cdcf6df4f328485971490d7024d : FAILURE in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/191ae653a04a442ea24b365b50f0894d : SUCCESS in 1h 44m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a18c68968fa4b9b91d3e7fa97b0615c : FAILURE in 2h 41m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8896210a7a3d4247a77a9a6f2c74b348 : SUCCESS in 2h 20m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8a84aadf860448d4a511b77a73f5ee58 : SUCCESS in 2h 08m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/78c952a60001418d9f27f62a80cd37cb : SUCCESS in 37m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/34e70d6238bb486692da241af0db61a2 : SUCCESS in 1h 19m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/90a7353a352e431d9d6d828934cc2825 : SUCCESS in 1h 50m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2a986011988141328feca5a036465242 : SUCCESS in 44m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9f919617f1884b558b568ee0e366597f : FAILURE in 2h 00m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3dbba311e3644f3a0ec20c02135fd82 : SUCCESS in 1h 53m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55538746886b4d05bf76514404062ccb : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":56},{"id":"125eda8929a8d6c222ab96f28462b78e2068a7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-04-30 15:08:44.000000000","message":"Uploaded patch set 57: Patch Set 56 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":57},{"id":"2ddbecaaa3605b65b07b924373d576dab02302c8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-30 16:18:32.000000000","message":"Patch Set 57:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I991e7481b0a9e031c303d60c5a5ab77fc257341c .","accounts_in_message":[],"_revision_number":57},{"id":"16228360bc3b4763b906448d6eced67c68ec78b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 16:48:45.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/619f93779d984e16b3b2a65492a97689\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f59bcf4f618148818c9ab2a10771070e : SUCCESS in 17m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4e1f9d354446405a9dcb027bfadbdfde : SUCCESS in 16m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/494e8030b6074f0aaf116173701d8eb2 : SUCCESS in 24m 19s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"f4e07c906d9851f422a700a626cabacc8d263a20","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-04-30 17:54:43.000000000","message":"Patch Set 57:\n\nBuild failed.\n\n- hpe-fc-driver build/1384a716d25e415fb6cfde072b66229a : FAILURE in 31m 07s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/958465/etc/tempest_log.log","accounts_in_message":[],"_revision_number":57},{"id":"cfc251da683145887d997acc5144e95da2bb3ec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 18:03:21.000000000","message":"Patch Set 57: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de755d8121754ae9ace4323db682ad2b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6ce1784f2b41f2b0d9012290fde1aa : SUCCESS in 6m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36b4a91dca6a410cbe867fa826bb716c : SUCCESS in 13m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8324b42f328b4eae95913685f5a1644a : SUCCESS in 10m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a7ea85a1de1944a6b3a19ad5ea4993d5 : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62a3c477aca34de7bb29f7c792e502c9 : SUCCESS in 15m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/a61b04171789460baf5757c2c491190f : SUCCESS in 1h 07m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/192f4e9360c5420d91b66895c74254a2 : SUCCESS in 46m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c2f388a92e4a4144a032c12581a85d99 : SUCCESS in 1h 35m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17faa4abd8df4dfa997f7352839cbc2d : SUCCESS in 1h 02m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/61d004cd6b124b309c0cfe84ab691423 : FAILURE in 25m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/78bc088491a144bd932141a99dd9f434 : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/992f96b48e7f45df97e6bf080ed43bfe : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7233af5c81774838bcae0391cb55448b : SUCCESS in 6m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/505da4457de544aabc0152a3ff843bfc : SUCCESS in 7m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2d86e7b327284569ac044361ea2b2b1a : SUCCESS in 40m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/21f37ba681c341f698196c740676102f : FAILURE in 8m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/548a56034b7545beae380eed87a87d93 : SUCCESS in 1h 55m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/698a898919994d6a895e35826ce0b690 : FAILURE in 1h 50m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df09e74d5e6244919f6f3bd2345b8b2c : SUCCESS in 2h 11m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/30501a5c046143ae8dc32e385bc29ee3 : SUCCESS in 2h 19m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f10114f750df4e03af32abd9b231ec83 : SUCCESS in 39m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91eb515c62ff49f9af973438c06dac88 : SUCCESS in 1h 19m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0139ec41eb6e4ee3a77f7989e42ed1e6 : SUCCESS in 1h 43m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2cfc50556fe442b492fe40d2334d3ceb : SUCCESS in 43m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5df49e1d79d34611a13ff597dbbb5801 : POST_FAILURE in 2h 41m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86844920553a43078d515da6b3018720 : SUCCESS in 1h 57m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c1a0545312f44c5b91549975efedd41 : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":57},{"id":"4ef133932ab255771dc0617587d91a9229b02b3f","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-30 20:38:27.000000000","message":"Patch Set 57:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/958465 : FAILURE","accounts_in_message":[],"_revision_number":57},{"id":"66b755f19a1316eabaf799cb012a705e498b66a9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-30 22:33:18.000000000","message":"Patch Set 57:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4027/.","accounts_in_message":[],"_revision_number":57},{"id":"eec85ff752dc8ea5f834a1faa3e1fa5521c3524c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-30 22:37:31.000000000","message":"Patch Set 57:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3851/.","accounts_in_message":[],"_revision_number":57},{"id":"82295d0cbbce0f1572e90bf04e3af5561cadcdb1","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-30 22:51:07.000000000","message":"Patch Set 57:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/958465 : SUCCESS","accounts_in_message":[],"_revision_number":57},{"id":"110fe90161887524ea1923310bc5ae3bd210b08e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-05-11 06:56:09.000000000","message":"Abandoned\n\nmearged changes in https://review.opendev.org/c/openstack/cinder/+/949006 patch","accounts_in_message":[],"_revision_number":57},{"id":"ff21ea347991991b0b5eba83183d03923e4bcad2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:56:22.000000000","message":"Patch Set 55:\n\nBuild Failed","accounts_in_message":[],"_revision_number":55},{"id":"7aedd5ee94c4974f8864cc463666bc810a613ee1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:56:30.000000000","message":"Patch Set 54:\n\nBuild Failed","accounts_in_message":[],"_revision_number":54},{"id":"50e0548443e0e08127afc99b6b59deb780e66857","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:57:26.000000000","message":"Patch Set 53:\n\nBuild Failed","accounts_in_message":[],"_revision_number":53},{"id":"61d231cb93c7c364ab2349688a066fcce2b9371f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:01:02.000000000","message":"Patch Set 57:\n\nBuild Failed","accounts_in_message":[],"_revision_number":57},{"id":"41bc7df7d0e092d3cf84dd8e822daf541937a3a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:01:11.000000000","message":"Patch Set 56:\n\nBuild Failed","accounts_in_message":[],"_revision_number":56}],"current_revision_number":57,"current_revision":"c0cf87cb23a7c1cf64b774b7310323881e40e0d7","revisions":{"8ec646a5c9a062c5a4878b1f34be7daeddb8f501":{"kind":"REWORK","_number":1,"created":"2025-08-25 18:10:54.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/1"}}},"commit":{"parents":[{"commit":"3c2cbd8314e12afe7b0f118885b006438947e269","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c2cbd8314e12afe7b0f118885b006438947e269"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 18:09:56.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ec646a5c9a062c5a4878b1f34be7daeddb8f501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ec646a5c9a062c5a4878b1f34be7daeddb8f501"}]},"branch":"refs/heads/master"},"2ba717b2a279cab52ae54961fa1b8f17ad218c70":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-25 18:12:13.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/2"}}},"commit":{"parents":[{"commit":"3c2cbd8314e12afe7b0f118885b006438947e269","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c2cbd8314e12afe7b0f118885b006438947e269"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 18:12:13.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.(VGS 2.0)","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.(VGS 2.0)\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ba717b2a279cab52ae54961fa1b8f17ad218c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ba717b2a279cab52ae54961fa1b8f17ad218c70"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e672aecc1c9ec518d8b47112b9d9b3eddf32c400":{"kind":"REWORK","_number":3,"created":"2025-08-25 18:13:12.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/3"}}},"commit":{"parents":[{"commit":"3c2cbd8314e12afe7b0f118885b006438947e269","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c2cbd8314e12afe7b0f118885b006438947e269"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 18:13:11.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.(VGS 2.0)","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.(VGS 2.0)\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e672aecc1c9ec518d8b47112b9d9b3eddf32c400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e672aecc1c9ec518d8b47112b9d9b3eddf32c400"}]},"branch":"refs/heads/master"},"8b4282d4559e4746c40937df21f0aefe3450cd33":{"kind":"REWORK","_number":4,"created":"2025-08-26 10:48:04.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/4"}}},"commit":{"parents":[{"commit":"3c2cbd8314e12afe7b0f118885b006438947e269","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c2cbd8314e12afe7b0f118885b006438947e269"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-26 10:47:19.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b4282d4559e4746c40937df21f0aefe3450cd33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b4282d4559e4746c40937df21f0aefe3450cd33"}]},"branch":"refs/heads/master"},"d8a3d5763284f1ab84b6eb0d309665be4d195c32":{"kind":"REWORK","_number":5,"created":"2025-08-26 10:51:28.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/5"}}},"commit":{"parents":[{"commit":"3c2cbd8314e12afe7b0f118885b006438947e269","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c2cbd8314e12afe7b0f118885b006438947e269"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-26 10:51:12.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8a3d5763284f1ab84b6eb0d309665be4d195c32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8a3d5763284f1ab84b6eb0d309665be4d195c32"}]},"branch":"refs/heads/master"},"6613e270cd788f0c9ae94f6c229af026ac8c7f53":{"kind":"REWORK","_number":6,"created":"2025-08-26 17:46:20.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/6"}}},"commit":{"parents":[{"commit":"3c2cbd8314e12afe7b0f118885b006438947e269","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c2cbd8314e12afe7b0f118885b006438947e269"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-26 17:45:54.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6613e270cd788f0c9ae94f6c229af026ac8c7f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6613e270cd788f0c9ae94f6c229af026ac8c7f53"}]},"branch":"refs/heads/master"},"9f0c85b212c89d8f43a08b0f9be9731096fa1b83":{"kind":"REWORK","_number":7,"created":"2025-09-01 06:51:59.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/7"}}},"commit":{"parents":[{"commit":"3c2cbd8314e12afe7b0f118885b006438947e269","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c2cbd8314e12afe7b0f118885b006438947e269"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-01 06:51:59.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9f0c85b212c89d8f43a08b0f9be9731096fa1b83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9f0c85b212c89d8f43a08b0f9be9731096fa1b83"}]},"branch":"refs/heads/master"},"8ee2760c695cb5c83d80080281d7990db71b244f":{"kind":"REWORK","_number":8,"created":"2025-09-01 06:52:40.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/8"}}},"commit":{"parents":[{"commit":"3c2cbd8314e12afe7b0f118885b006438947e269","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c2cbd8314e12afe7b0f118885b006438947e269"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-01 06:52:39.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ee2760c695cb5c83d80080281d7990db71b244f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ee2760c695cb5c83d80080281d7990db71b244f"}]},"branch":"refs/heads/master"},"43209071ae7b0323ae98d62aac525c9d4cf2399a":{"kind":"REWORK","_number":9,"created":"2025-09-01 06:53:38.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/9"}}},"commit":{"parents":[{"commit":"3c2cbd8314e12afe7b0f118885b006438947e269","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c2cbd8314e12afe7b0f118885b006438947e269"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-01 06:53:37.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43209071ae7b0323ae98d62aac525c9d4cf2399a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43209071ae7b0323ae98d62aac525c9d4cf2399a"}]},"branch":"refs/heads/master"},"3e19d02f977918604534618e56328f039ece82bd":{"kind":"REWORK","_number":10,"created":"2025-09-01 07:16:01.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/10"}}},"commit":{"parents":[{"commit":"3c2cbd8314e12afe7b0f118885b006438947e269","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c2cbd8314e12afe7b0f118885b006438947e269"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-01 07:16:01.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e19d02f977918604534618e56328f039ece82bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e19d02f977918604534618e56328f039ece82bd"}]},"branch":"refs/heads/master"},"e4cf0dd6d31a051b409d2afaff1ec46c0fb049e4":{"kind":"REWORK","_number":11,"created":"2025-09-01 07:16:38.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/11"}}},"commit":{"parents":[{"commit":"3c2cbd8314e12afe7b0f118885b006438947e269","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c2cbd8314e12afe7b0f118885b006438947e269"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-01 07:16:38.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e4cf0dd6d31a051b409d2afaff1ec46c0fb049e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e4cf0dd6d31a051b409d2afaff1ec46c0fb049e4"}]},"branch":"refs/heads/master"},"3e1123ab56ad507789ed562f4e59d37478cb7ebd":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-09-03 05:59:35.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/12"}}},"commit":{"parents":[{"commit":"65646bd731c45a5786c603272f1654c4c7f5ba87","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65646bd731c45a5786c603272f1654c4c7f5ba87"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-03 05:59:35.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e1123ab56ad507789ed562f4e59d37478cb7ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e1123ab56ad507789ed562f4e59d37478cb7ebd"}]},"branch":"refs/heads/master","description":"Rebase"},"1aa2bd699221cebfc86ca11c9ce8b29c1cf73506":{"kind":"REWORK","_number":13,"created":"2025-09-03 06:11:02.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/13"}}},"commit":{"parents":[{"commit":"ce402df10789973916bbf855546bf9e857d6a2c6","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce402df10789973916bbf855546bf9e857d6a2c6"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-03 06:10:27.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aa2bd699221cebfc86ca11c9ce8b29c1cf73506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aa2bd699221cebfc86ca11c9ce8b29c1cf73506"}]},"branch":"refs/heads/master"},"a4d419f84247e9d0cd6d623cb38432ea6ba49b69":{"kind":"REWORK","_number":14,"created":"2025-09-03 06:56:15.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/14"}}},"commit":{"parents":[{"commit":"ce402df10789973916bbf855546bf9e857d6a2c6","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce402df10789973916bbf855546bf9e857d6a2c6"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-03 06:56:13.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4d419f84247e9d0cd6d623cb38432ea6ba49b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4d419f84247e9d0cd6d623cb38432ea6ba49b69"}]},"branch":"refs/heads/master"},"e99f3f34d94c667480fb85d0e27e561bb4e3287f":{"kind":"REWORK","_number":15,"created":"2025-09-03 06:57:51.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/15"}}},"commit":{"parents":[{"commit":"ce402df10789973916bbf855546bf9e857d6a2c6","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce402df10789973916bbf855546bf9e857d6a2c6"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-03 06:57:50.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e99f3f34d94c667480fb85d0e27e561bb4e3287f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e99f3f34d94c667480fb85d0e27e561bb4e3287f"}]},"branch":"refs/heads/master"},"b085a6066dc38afaab8ce159304a9e6408aca853":{"kind":"REWORK","_number":16,"created":"2025-09-03 06:58:32.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/16"}}},"commit":{"parents":[{"commit":"ce402df10789973916bbf855546bf9e857d6a2c6","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce402df10789973916bbf855546bf9e857d6a2c6"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-03 06:58:31.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b085a6066dc38afaab8ce159304a9e6408aca853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b085a6066dc38afaab8ce159304a9e6408aca853"}]},"branch":"refs/heads/master"},"74c3e073c94c94a80334f43e224a60c5b2a0b50f":{"kind":"REWORK","_number":17,"created":"2025-09-03 06:59:27.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/17"}}},"commit":{"parents":[{"commit":"ce402df10789973916bbf855546bf9e857d6a2c6","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce402df10789973916bbf855546bf9e857d6a2c6"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-03 06:59:27.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74c3e073c94c94a80334f43e224a60c5b2a0b50f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74c3e073c94c94a80334f43e224a60c5b2a0b50f"}]},"branch":"refs/heads/master"},"2bbf0c146943e2dd1a240123ccdc929ab6313740":{"kind":"REWORK","_number":18,"created":"2025-09-03 07:32:35.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/18"}}},"commit":{"parents":[{"commit":"ce402df10789973916bbf855546bf9e857d6a2c6","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce402df10789973916bbf855546bf9e857d6a2c6"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-03 07:32:35.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bbf0c146943e2dd1a240123ccdc929ab6313740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bbf0c146943e2dd1a240123ccdc929ab6313740"}]},"branch":"refs/heads/master"},"75e6108a1b908158a0a6026955eb51f74c8f0fd6":{"kind":"REWORK","_number":19,"created":"2025-09-03 07:33:04.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/19"}}},"commit":{"parents":[{"commit":"ce402df10789973916bbf855546bf9e857d6a2c6","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce402df10789973916bbf855546bf9e857d6a2c6"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-03 07:33:04.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75e6108a1b908158a0a6026955eb51f74c8f0fd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75e6108a1b908158a0a6026955eb51f74c8f0fd6"}]},"branch":"refs/heads/master"},"bf9a72a3c6d42cfc961c3f029a5d81e5b3403469":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2025-09-08 10:47:16.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/20"}}},"commit":{"parents":[{"commit":"9274b73c2e2d20c1c8cdb0ad304e4905c7e36bfc","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9274b73c2e2d20c1c8cdb0ad304e4905c7e36bfc"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-08 10:47:16.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf9a72a3c6d42cfc961c3f029a5d81e5b3403469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf9a72a3c6d42cfc961c3f029a5d81e5b3403469"}]},"branch":"refs/heads/master","description":"Rebase"},"3d521f68485cc925d2a9a2206d28f54a0d90359c":{"kind":"REWORK","_number":21,"created":"2025-09-10 15:20:15.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/21"}}},"commit":{"parents":[{"commit":"0e675d13992208d47291882602b5753ad482b959","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e675d13992208d47291882602b5753ad482b959"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-09-10 15:19:20.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d521f68485cc925d2a9a2206d28f54a0d90359c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d521f68485cc925d2a9a2206d28f54a0d90359c"}]},"branch":"refs/heads/master"},"c94fb26cddf6f9bb4cc942dfea442b5363423099":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2025-10-31 15:32:34.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/22"}}},"commit":{"parents":[{"commit":"8abd1d1353129835ccc864bdaefa7514af1c841e","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8abd1d1353129835ccc864bdaefa7514af1c841e"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"dakshinapurohit","email":"Dakshina.Purohit@ibm.com","date":"2025-10-31 20:56:06.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: Dakshina Purohit \u003cdakshina.purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c94fb26cddf6f9bb4cc942dfea442b5363423099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c94fb26cddf6f9bb4cc942dfea442b5363423099"}]},"branch":"refs/heads/master"},"18bac848887c013b23b8384ab51f1b9a98d79f77":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":23,"created":"2025-11-05 13:47:46.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/23"}}},"commit":{"parents":[{"commit":"8927088a4602aecd8e982527d0fedf3f79aa5030","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8927088a4602aecd8e982527d0fedf3f79aa5030"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-11-05 13:47:05.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18bac848887c013b23b8384ab51f1b9a98d79f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18bac848887c013b23b8384ab51f1b9a98d79f77"}]},"branch":"refs/heads/master"},"20240d10ec75643844cc0ef3428c3f7971540c15":{"kind":"REWORK","_number":24,"created":"2025-11-07 11:46:00.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/24"}}},"commit":{"parents":[{"commit":"abefdf0c6d5fc6a3eb7ed59e0ba0164ef878e889","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/abefdf0c6d5fc6a3eb7ed59e0ba0164ef878e889"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-11-07 11:45:15.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20240d10ec75643844cc0ef3428c3f7971540c15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20240d10ec75643844cc0ef3428c3f7971540c15"}]},"branch":"refs/heads/master"},"82a65073f3b78827f7dcd77992e97b5d8eb4f8fc":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2025-11-11 10:20:43.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/65/958465/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/25"}}},"commit":{"parents":[{"commit":"185650e8ef485f28bf0992d83fac5fe526235733","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/185650e8ef485f28bf0992d83fac5fe526235733"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2025-11-11 10:15:25.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/82a65073f3b78827f7dcd77992e97b5d8eb4f8fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/82a65073f3b78827f7dcd77992e97b5d8eb4f8fc"}]},"branch":"refs/heads/master"},"d6b4347a80f0c75b5462b050cecf1c7b11937a0d":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2025-11-12 09:35:16.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/26"}}},"commit":{"parents":[{"commit":"118659f572713e6db0088aab08050b4fdce61266","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/118659f572713e6db0088aab08050b4fdce61266"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-11-12 05:27:19.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6b4347a80f0c75b5462b050cecf1c7b11937a0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6b4347a80f0c75b5462b050cecf1c7b11937a0d"}]},"branch":"refs/heads/master"},"8dc767f39780e6c459c38f1d5b4c5f55947cb4f0":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2025-11-13 10:22:02.000000000","uploader":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"ref":"refs/changes/65/958465/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/27"}}},"commit":{"parents":[{"commit":"394504b61f45182f5a321109df0dbcdba54fcfa2","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/394504b61f45182f5a321109df0dbcdba54fcfa2"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"rakshay","email":"akshay.ranade@ibm.com","date":"2025-11-13 10:19:26.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8dc767f39780e6c459c38f1d5b4c5f55947cb4f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8dc767f39780e6c459c38f1d5b4c5f55947cb4f0"}]},"branch":"refs/heads/master"},"888066304068d69572962842ec87d83d92337d7d":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2025-11-13 11:52:09.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/28"}}},"commit":{"parents":[{"commit":"c1a44182655ac7c871e60959d88a1ae5212b5ed1","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1a44182655ac7c871e60959d88a1ae5212b5ed1"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-11-13 05:51:00.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/888066304068d69572962842ec87d83d92337d7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/888066304068d69572962842ec87d83d92337d7d"}]},"branch":"refs/heads/master"},"638673d94150c80972ac1ef0981cdc86e371f8fe":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2025-11-13 11:57:28.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/29"}}},"commit":{"parents":[{"commit":"5d786edb4aae1f7c67aa28626b085111fc17d313","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d786edb4aae1f7c67aa28626b085111fc17d313"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"dakshinapurohit","email":"Dakshina.Purohit@ibm.com","date":"2025-11-13 16:50:47.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/638673d94150c80972ac1ef0981cdc86e371f8fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/638673d94150c80972ac1ef0981cdc86e371f8fe"}]},"branch":"refs/heads/master"},"c0bf5029409a4eb5542919395c615f4928874c18":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2025-11-20 05:54:45.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/65/958465/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/30"}}},"commit":{"parents":[{"commit":"e3ab5d5a83c638d263b9b7104d6617dbe090fa23","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3ab5d5a83c638d263b9b7104d6617dbe090fa23"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2025-11-20 05:34:56.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0bf5029409a4eb5542919395c615f4928874c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0bf5029409a4eb5542919395c615f4928874c18"}]},"branch":"refs/heads/master"},"579c1b3146458eb2ae722d43909f9a676bc2afba":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2026-01-05 08:30:10.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/31"}}},"commit":{"parents":[{"commit":"ad15e0fbfd918336a3e82de5ab7f6da97fa0287f","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad15e0fbfd918336a3e82de5ab7f6da97fa0287f"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-12-24 08:46:48.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/579c1b3146458eb2ae722d43909f9a676bc2afba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/579c1b3146458eb2ae722d43909f9a676bc2afba"}]},"branch":"refs/heads/master"},"6207d47881abc2979a9e831ab41d37d64b37a09d":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2026-01-06 10:23:31.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/65/958465/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/32"}}},"commit":{"parents":[{"commit":"96edec89c4713d55456c5ab59feb5a10dca7a9ae","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96edec89c4713d55456c5ab59feb5a10dca7a9ae"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-01-06 09:13:07.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6207d47881abc2979a9e831ab41d37d64b37a09d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6207d47881abc2979a9e831ab41d37d64b37a09d"}]},"branch":"refs/heads/master"},"98ed6ac47c11ed0eeb0bed76503bdd5db66b12dd":{"kind":"REWORK","_number":33,"created":"2026-01-07 14:37:34.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/33"}}},"commit":{"parents":[{"commit":"96edec89c4713d55456c5ab59feb5a10dca7a9ae","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96edec89c4713d55456c5ab59feb5a10dca7a9ae"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-01-07 14:37:33.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98ed6ac47c11ed0eeb0bed76503bdd5db66b12dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98ed6ac47c11ed0eeb0bed76503bdd5db66b12dd"}]},"branch":"refs/heads/master"},"86c881aea3ef169cafadc55847cc0bef54c825c9":{"kind":"REWORK","_number":34,"created":"2026-01-07 14:37:59.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/34"}}},"commit":{"parents":[{"commit":"96edec89c4713d55456c5ab59feb5a10dca7a9ae","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96edec89c4713d55456c5ab59feb5a10dca7a9ae"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-01-07 14:37:59.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/86c881aea3ef169cafadc55847cc0bef54c825c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/86c881aea3ef169cafadc55847cc0bef54c825c9"}]},"branch":"refs/heads/master"},"1055645ccd9862ef1a352729ce53a6e81f2a178e":{"kind":"REWORK","_number":35,"created":"2026-01-07 14:39:16.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/35"}}},"commit":{"parents":[{"commit":"96edec89c4713d55456c5ab59feb5a10dca7a9ae","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96edec89c4713d55456c5ab59feb5a10dca7a9ae"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-01-07 14:39:15.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1055645ccd9862ef1a352729ce53a6e81f2a178e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1055645ccd9862ef1a352729ce53a6e81f2a178e"}]},"branch":"refs/heads/master"},"50d86e7b5f6d0848d7b43ccbb81e82c9bedf1267":{"kind":"REWORK","_number":36,"created":"2026-01-07 14:39:40.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/36"}}},"commit":{"parents":[{"commit":"96edec89c4713d55456c5ab59feb5a10dca7a9ae","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96edec89c4713d55456c5ab59feb5a10dca7a9ae"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-01-07 14:39:39.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50d86e7b5f6d0848d7b43ccbb81e82c9bedf1267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50d86e7b5f6d0848d7b43ccbb81e82c9bedf1267"}]},"branch":"refs/heads/master"},"73fc04713ff75996e8069f16cbe1d666579b01c6":{"kind":"REWORK","_number":37,"created":"2026-01-07 17:58:02.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/37"}}},"commit":{"parents":[{"commit":"96edec89c4713d55456c5ab59feb5a10dca7a9ae","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96edec89c4713d55456c5ab59feb5a10dca7a9ae"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-01-07 17:58:01.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/73fc04713ff75996e8069f16cbe1d666579b01c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/73fc04713ff75996e8069f16cbe1d666579b01c6"}]},"branch":"refs/heads/master"},"357ae61c1f2ceb5a2f62ed109dd3b803ef2bde57":{"kind":"REWORK","_number":38,"created":"2026-01-07 18:00:00.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/38"}}},"commit":{"parents":[{"commit":"96edec89c4713d55456c5ab59feb5a10dca7a9ae","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96edec89c4713d55456c5ab59feb5a10dca7a9ae"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-01-07 17:59:56.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/357ae61c1f2ceb5a2f62ed109dd3b803ef2bde57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/357ae61c1f2ceb5a2f62ed109dd3b803ef2bde57"}]},"branch":"refs/heads/master"},"f46412f401023dc22995e2c6b08c42744990ff29":{"kind":"REWORK","_number":39,"created":"2026-01-14 08:17:29.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/39"}}},"commit":{"parents":[{"commit":"96edec89c4713d55456c5ab59feb5a10dca7a9ae","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96edec89c4713d55456c5ab59feb5a10dca7a9ae"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-01-14 08:17:29.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f46412f401023dc22995e2c6b08c42744990ff29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f46412f401023dc22995e2c6b08c42744990ff29"}]},"branch":"refs/heads/master"},"23713d62d679cdcc6757a300157f2cb09566fdc1":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2026-01-16 03:33:51.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"real_uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/40"}}},"commit":{"parents":[{"commit":"607fae499505d1de7d8e69139d91cf2c0d764516","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/607fae499505d1de7d8e69139d91cf2c0d764516"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-01-16 03:33:51.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23713d62d679cdcc6757a300157f2cb09566fdc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23713d62d679cdcc6757a300157f2cb09566fdc1"}]},"branch":"refs/heads/master","description":"Rebase"},"623f37681b76cd8f849f4a18ad129e80e118a0a9":{"kind":"NO_CHANGE","_number":41,"created":"2026-01-20 05:52:50.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/65/958465/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/41"}}},"commit":{"parents":[{"commit":"0942d6951f0182226e04bb497e0ddc0390bf4bba","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0942d6951f0182226e04bb497e0ddc0390bf4bba"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-19 05:52:35.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/623f37681b76cd8f849f4a18ad129e80e118a0a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/623f37681b76cd8f849f4a18ad129e80e118a0a9"}]},"branch":"refs/heads/master"},"8b3f0964bc52607e33c7838a39219f11a8ab830d":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2026-01-28 06:59:48.000000000","uploader":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"ref":"refs/changes/65/958465/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/42"}}},"commit":{"parents":[{"commit":"0ff9c6aad803e73747a8138429d33757dfa267a8","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ff9c6aad803e73747a8138429d33757dfa267a8"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Sandip27","email":"sandip.rajbanshi@ibm.com","date":"2026-01-27 10:57:43.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b3f0964bc52607e33c7838a39219f11a8ab830d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b3f0964bc52607e33c7838a39219f11a8ab830d"}]},"branch":"refs/heads/master"},"502fc912984cab535e51a41a48ce83d8c50600fe":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2026-01-29 11:51:22.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/43"}}},"commit":{"parents":[{"commit":"7ac5dfc4d11a0384596cae1ca78a67e84add6436","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ac5dfc4d11a0384596cae1ca78a67e84add6436"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-01-29 11:45:33.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/502fc912984cab535e51a41a48ce83d8c50600fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/502fc912984cab535e51a41a48ce83d8c50600fe"}]},"branch":"refs/heads/master"},"71b6fa8b111f4b060d8b8a95f821d6a84f6f6dbb":{"kind":"REWORK","_number":44,"created":"2026-02-03 08:44:17.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/44"}}},"commit":{"parents":[{"commit":"7ac5dfc4d11a0384596cae1ca78a67e84add6436","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ac5dfc4d11a0384596cae1ca78a67e84add6436"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-02-03 08:43:54.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71b6fa8b111f4b060d8b8a95f821d6a84f6f6dbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71b6fa8b111f4b060d8b8a95f821d6a84f6f6dbb"}]},"branch":"refs/heads/master"},"55a9175588f5d9fb7de8d6911542109e6010a22b":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2026-02-05 06:38:56.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/45"}}},"commit":{"parents":[{"commit":"72e638dccfb10cf49c0eeeb7ac500b6d84e02819","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72e638dccfb10cf49c0eeeb7ac500b6d84e02819"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-02-05 06:33:54.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55a9175588f5d9fb7de8d6911542109e6010a22b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55a9175588f5d9fb7de8d6911542109e6010a22b"}]},"branch":"refs/heads/master"},"5ac0a3a5824d12a04bd1600d5944fdfb4820cc35":{"kind":"REWORK","_number":46,"created":"2026-02-05 06:52:25.000000000","uploader":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"ref":"refs/changes/65/958465/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/46"}}},"commit":{"parents":[{"commit":"f063b9be7d7d2beab779f219d66021a10164409e","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f063b9be7d7d2beab779f219d66021a10164409e"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"sandip-rajbanshi-27","email":"sandip.rajbanshi@ibm.com","date":"2026-02-05 06:50:40.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ac0a3a5824d12a04bd1600d5944fdfb4820cc35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ac0a3a5824d12a04bd1600d5944fdfb4820cc35"}]},"branch":"refs/heads/master"},"9bb2e729ea5b700412daa55638704b3f70b5c4be":{"kind":"TRIVIAL_REBASE","_number":47,"created":"2026-02-05 09:44:17.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/65/958465/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/47"}}},"commit":{"parents":[{"commit":"28d21e5836a2d8e41ba43815bccbdbb071b0cb61","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28d21e5836a2d8e41ba43815bccbdbb071b0cb61"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-02-05 09:34:38.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9bb2e729ea5b700412daa55638704b3f70b5c4be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9bb2e729ea5b700412daa55638704b3f70b5c4be"}]},"branch":"refs/heads/master"},"bfdf7e36244af3ba6b9e327e03b7153fdaed11d5":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2026-02-13 09:22:24.000000000","uploader":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"ref":"refs/changes/65/958465/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/48"}}},"commit":{"parents":[{"commit":"5353a680756e06f1d7c70936ea4a74a85e300786","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5353a680756e06f1d7c70936ea4a74a85e300786"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"sandip-rajbanshi-27","email":"sandip.rajbanshi@ibm.com","date":"2026-02-13 07:39:08.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bfdf7e36244af3ba6b9e327e03b7153fdaed11d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bfdf7e36244af3ba6b9e327e03b7153fdaed11d5"}]},"branch":"refs/heads/master"},"d4694ce9484c72240dcdb76e257a6d97d012f6af":{"kind":"TRIVIAL_REBASE","_number":49,"created":"2026-03-12 05:55:33.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/49"}}},"commit":{"parents":[{"commit":"e86e330cd4123bcfaffac6b169061bc3a256bcb3","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e86e330cd4123bcfaffac6b169061bc3a256bcb3"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-03-12 05:53:13.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4694ce9484c72240dcdb76e257a6d97d012f6af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4694ce9484c72240dcdb76e257a6d97d012f6af"}]},"branch":"refs/heads/master"},"04bcfa6d8589182043f0223164aaebb20e4a4ed9":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2026-03-17 10:55:03.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/65/958465/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/50"}}},"commit":{"parents":[{"commit":"6b8892e80252bd40b667105f1492ea8c6c5672fc","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6b8892e80252bd40b667105f1492ea8c6c5672fc"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-03-17 10:50:13.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04bcfa6d8589182043f0223164aaebb20e4a4ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04bcfa6d8589182043f0223164aaebb20e4a4ed9"}]},"branch":"refs/heads/master"},"908940d6bc7e1f9dfdebf79d048e0c4f3c0ee85e":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2026-03-31 05:34:39.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/65/958465/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/51"}}},"commit":{"parents":[{"commit":"303f394561b35e105420571197faba393bb3e60b","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/303f394561b35e105420571197faba393bb3e60b"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-03-31 04:37:04.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/908940d6bc7e1f9dfdebf79d048e0c4f3c0ee85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/908940d6bc7e1f9dfdebf79d048e0c4f3c0ee85e"}]},"branch":"refs/heads/master"},"f6e3f720d65274f02be4724714fe08c3cea3f5ab":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2026-03-31 09:40:01.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"real_uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/65/958465/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/52"}}},"commit":{"parents":[{"commit":"7a2042b6a491968b97199ab8ad8e783e81ece3c7","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7a2042b6a491968b97199ab8ad8e783e81ece3c7"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Vrinda Dhakad","email":"vrinda.dhakad@ibm.com","date":"2026-03-31 09:40:01.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6e3f720d65274f02be4724714fe08c3cea3f5ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6e3f720d65274f02be4724714fe08c3cea3f5ab"}]},"branch":"refs/heads/master","description":"Rebase"},"6496be5ca54b73b9593ed7c9ae6c5dbceb2c4474":{"kind":"TRIVIAL_REBASE","_number":53,"created":"2026-04-09 10:19:31.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/65/958465/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/53"}}},"commit":{"parents":[{"commit":"d85a64ccff0d773085502bc39869cc55769e5850","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d85a64ccff0d773085502bc39869cc55769e5850"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-04-09 06:23:23.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6496be5ca54b73b9593ed7c9ae6c5dbceb2c4474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6496be5ca54b73b9593ed7c9ae6c5dbceb2c4474"}]},"branch":"refs/heads/master"},"f493409bdcac33c2b669f3ef62ea64ec6d156beb":{"kind":"NO_CHANGE","_number":54,"created":"2026-04-13 07:27:00.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/54"}}},"commit":{"parents":[{"commit":"3cfd3445948f4aee96d0c92f83795c1c3e11ee51","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3cfd3445948f4aee96d0c92f83795c1c3e11ee51"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-04-13 05:58:52.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f493409bdcac33c2b669f3ef62ea64ec6d156beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f493409bdcac33c2b669f3ef62ea64ec6d156beb"}]},"branch":"refs/heads/master"},"f4831554b14854d20bb21c13490dd5771e38976d":{"kind":"NO_CHANGE","_number":55,"created":"2026-04-14 11:04:51.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/55"}}},"commit":{"parents":[{"commit":"6784d452f17df3ce04dcdf41653837f386297749","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6784d452f17df3ce04dcdf41653837f386297749"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-04-14 11:00:42.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4831554b14854d20bb21c13490dd5771e38976d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4831554b14854d20bb21c13490dd5771e38976d"}]},"branch":"refs/heads/master"},"4ff3401dba0ea9350c68acb718517c0f42d42269":{"kind":"TRIVIAL_REBASE","_number":56,"created":"2026-04-28 17:53:21.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/65/958465/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/56"}}},"commit":{"parents":[{"commit":"812818dca2e55e0e2008f742f188162a93d84d38","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/812818dca2e55e0e2008f742f188162a93d84d38"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-04-28 17:22:41.000000000","tz":330},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ff3401dba0ea9350c68acb718517c0f42d42269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ff3401dba0ea9350c68acb718517c0f42d42269"}]},"branch":"refs/heads/master"},"c0cf87cb23a7c1cf64b774b7310323881e40e0d7":{"kind":"TRIVIAL_REBASE","_number":57,"created":"2026-04-30 15:08:44.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/65/958465/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/958465/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/958465/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/958465/57"}}},"commit":{"parents":[{"commit":"51e5f85d8bf513a12a366e48ede6d86ec68ddbd7","subject":"[IBM SVf driver] Adding support for Individual volume clone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51e5f85d8bf513a12a366e48ede6d86ec68ddbd7"}]}],"author":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2025-08-25 17:54:14.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-04-27 18:40:54.000000000","tz":0},"subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","message":"Doc: Documentation updates for ibm-svf-volumegroup feature.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: I991e7481b0a9e031c303d60c5a5ab77fc257341c\nSigned-off-by: dakshinapurohit \u003cDakshina.Purohit@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0cf87cb23a7c1cf64b774b7310323881e40e0d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0cf87cb23a7c1cf64b774b7310323881e40e0d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
