)]}'
{"id":"openstack%2Fcinder~958496","triplet_id":"openstack%2Fcinder~master~Ic4b204402fec482d632496f31ef9a2686ffc5d09","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"37923":{"account":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"last_update":"2025-08-26 17:19:06.000000000","reason":"Change was abandoned"},"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-08-26 06:56:09.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic4b204402fec482d632496f31ef9a2686ffc5d09","subject":"Hitachi: add support for Hitachi VSP One B20","status":"ABANDONED","created":"2025-08-25 22:42:54.000000000","updated":"2025-08-26 17:19:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2f9436ebe734ea954e01e373160e8960a5dc268f","_number":958496,"virtual_id_number":958496,"owner":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-26 02:46:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-25 22:44:00.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-08-25 22:45:33.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-25 23:29:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-26 02:46:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 05:04:17.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-26 05:31:13.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-26 05:40:26.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-26 11:42:16.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"}],"messages":[{"id":"675260df8eaf9c4f00baebe0c8bd0df665ee79e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-25 22:42:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d15444925a449c4e9f931abb2daa40eeba8fa877","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-25 22:44:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"304b1bde0017915d122e50d16a0ba1de25c7bc32","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-25 22:45:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a4813cd092d34f7ab1e42a98318684ce : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/deef90f9ed3149e29641455c363e883a : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"384d307375b10bddcb8ab85c1e1966bc559346c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-25 22:58:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12ff101174e1f79c5d72215ffda32da1cb313a7b","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-25 22:59:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"2e379478b5d7372c0d69666e988e9e0462eba6b8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-25 23:01:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9f7b9f91bb1645dc831a4f9c09c3a14d : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/03aca57f20a74b7f963237989bca328c : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"b97ffa7cd8a34f5f278c252bce82bba0807f7cd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 23:29:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5c17e6388a1450aa687743ae1e67daf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc74765afd2f418994844fce8873c5cf : SUCCESS in 22m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84440a92c48d4101a3810004606b0421 : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b01648bac0574f8b811a7a1b5ddd2609 : SUCCESS in 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9bf485a65c3ea534973c76c187c21d017bdb473f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-26 00:14:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1a81f805df556a06bc38826e69ddc219850ff7e0","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-26 00:15:15.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver hpe-nvmetcp-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"d16dd47b23855bbb359d36a8a4023fea7ca25858","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-26 00:17:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/25b5fab533754727afa36331d9d323e0 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/467cef4658054123836bd333e239f44a : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"4b63677ad8547f31206db11b00df25bf8b9db300","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 00:41:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f92b8fbb960c41049a93b3cb335731c5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e873e36690624133b622e0dcab42cd9b : FAILURE in 17m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e766a8b89b44c52a9f2a2c97bc97622 : FAILURE in 16m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eb49dad55bd74c55ae6005ca074e549b : FAILURE in 25m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6d34e164a290a44d91c4006c3d892c7ddd09c169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 02:46:24.000000000","message":"Patch Set 3: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/370e80bdd51a46a1b104853905bd70bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05ee634868d74828a2b94d5e0c13b616 : FAILURE in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac427297e4614a9a824a251324fc543a : FAILURE in 10m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2af9bc1641e14f92b40e8fe14264ecda : FAILURE in 9m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de59551a1fd847c48b0c6b477e090005 : FAILURE in 12m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3259db598c4b4728b7ce2e6afdb1c1db : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/7ea8a02313ed46b687cc0389f0bb53e6 : SUCCESS in 37m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9508efcfe4a14af9b296efd491f68e85 : SUCCESS in 1h 02m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8552d2473f1e47f6ad78db44aa5b5e38 : SUCCESS in 1h 32m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9fb6f71c5c844c7a4457633fac8c1c0 : SUCCESS in 56m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d9e59339304c4694b2a4d1e5a7d95410 : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/18e5e24c99504f14a8f57e009a1788f4 : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3dcdfaccbb764d8299f193deefc5a36f : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c4ab164200c04ae494a13ffb5261a53e : SUCCESS in 6m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2d99c6e3554042ed92ef40231e82fa28 : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/114224d5fdc54795a548708e5630b717 : FAILURE in 39m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cfa708ed0980407b97a9bba94c3bea17 : SUCCESS in 7m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd6d54b97901470b8059e820241450ec : SUCCESS in 2h 04m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3f7325cf934742b6af946b473c2b01d6 : FAILURE in 2h 23m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/33dd26dff7d2441981c2cc508080dd3f : SUCCESS in 2h 16m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f499e99843a64aa2a75ffd8c6dca48e1 : FAILURE in 21m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/87d257f1feb54c289ae97f8a8eae548a : SUCCESS in 17m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/04f5030887ba4939ba94a761cf9a1a49 : SUCCESS in 43m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/55dca0f3cbfb4a6ea4a85d9fd6ff7cbb : SUCCESS in 1h 36m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/83dc55ebc5c24fad856e09b47a7bff00 : SUCCESS in 22m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/77e1dcc14b3345b4ab9cb534186755d7 : SUCCESS in 1h 30m 40s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ce2d998d97cc4c13bd7d50bee46a8861 : FAILURE in 17m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a9e136d18284fcdb86b3bd095ae4d4f : SUCCESS in 49m 54s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b302f46454246f4a02308a7da5fa83d : SUCCESS in 1h 16m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bf64f0d7d374bcaa4cbe998766a3912 : SUCCESS in 28m 38s","accounts_in_message":[],"_revision_number":3},{"id":"4eeaf90993aaf24dae6698e4f8d7c315905bee71","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 05:04:17.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/298/","accounts_in_message":[],"_revision_number":1},{"id":"61fc13e02053912fd51f731dd7bf59fff674c79f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-26 05:31:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958496 : FAILURE in 37m 58s","accounts_in_message":[],"_revision_number":3},{"id":"1cbc5864827f46706f6cbd6f295f99393d308b62","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 05:40:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1697/.","accounts_in_message":[],"_revision_number":3},{"id":"1ff69e58f9a8600c5673957c69745e6b71c32851","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 05:59:37.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958496/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958496/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"954462d407a0a1f7eb372fec66e341d98b706d24","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 05:59:38.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/299/","accounts_in_message":[],"_revision_number":2},{"id":"8be137aaa1e43c53baf7b969395cea9baa6cda5f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 06:56:09.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958496/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958496/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"e86c00c26e385119415e9e0e919c5b7946cb48ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 06:56:11.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/300/","accounts_in_message":[],"_revision_number":3},{"id":"87ae05069c9874312767589f5faf47610e13faf1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 07:49:49.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958496/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958496/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"4f260b0d6f43c7a385dfc96b5c434e37eb0324fe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 08:23:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1563/.","accounts_in_message":[],"_revision_number":3},{"id":"6621f1b94bc1dca4c48c81249f2709a1336c9509","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-26 11:42:16.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-96-958496-2/FC/ : UNSTABLE in 2h 44m 36s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-96-958496-2/iSCSI/ : FAILURE in 6h 40m 15s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"5e669cb04b095699e66eb26cd554973fd34ff0f0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-26 12:56:00.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-96-958496-3/FC/ : FAILURE in 7h 7m 11s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-96-958496-3/iSCSI/ : FAILURE in 6h 19m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"2f9436ebe734ea954e01e373160e8960a5dc268f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-26 17:19:06.000000000","message":"Abandoned\n\ndidn\u0027t rebase to master","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b59053a000c74b22004ae93b75c508da57e95f33","revisions":{"8a54911dc998226fe5957260221c0cb1f686a12f":{"kind":"REWORK","_number":1,"created":"2025-08-25 22:42:54.000000000","uploader":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"ref":"refs/changes/96/958496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/96/958496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/958496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/958496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/958496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/96/958496/1"}}},"commit":{"parents":[{"commit":"2466ef31093b16e8080648744ccf5040bc18e54a","subject":"Hitachi: add support for Hitachi VSP One B20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2466ef31093b16e8080648744ccf5040bc18e54a"}]}],"author":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-25 22:42:24.000000000","tz":-420},"committer":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-25 22:42:46.000000000","tz":-420},"subject":"Hitachi: add support for Hitachi VSP One B20","message":"Hitachi: add support for Hitachi VSP One B20\n\nChange-Id: Ic4b204402fec482d632496f31ef9a2686ffc5d09\nSigned-off-by: Anthony Gamboa \u003canthony.gamboa@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a54911dc998226fe5957260221c0cb1f686a12f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a54911dc998226fe5957260221c0cb1f686a12f"}]},"branch":"refs/heads/master"},"7747ca8d5ed16a4964d354ead7939e2d2099f939":{"kind":"REWORK","_number":2,"created":"2025-08-25 22:58:28.000000000","uploader":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"ref":"refs/changes/96/958496/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/96/958496/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/958496/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/958496/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/958496/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/96/958496/2"}}},"commit":{"parents":[{"commit":"2466ef31093b16e8080648744ccf5040bc18e54a","subject":"Hitachi: add support for Hitachi VSP One B20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2466ef31093b16e8080648744ccf5040bc18e54a"}]}],"author":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-25 22:42:24.000000000","tz":-420},"committer":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-25 22:58:18.000000000","tz":-420},"subject":"Hitachi: add support for Hitachi VSP One B20","message":"Hitachi: add support for Hitachi VSP One B20\n\nChange-Id: Ic4b204402fec482d632496f31ef9a2686ffc5d09\nSigned-off-by: Anthony Gamboa \u003canthony.gamboa@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7747ca8d5ed16a4964d354ead7939e2d2099f939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7747ca8d5ed16a4964d354ead7939e2d2099f939"}]},"branch":"refs/heads/master"},"b59053a000c74b22004ae93b75c508da57e95f33":{"kind":"REWORK","_number":3,"created":"2025-08-26 00:14:09.000000000","uploader":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"ref":"refs/changes/96/958496/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/96/958496/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/958496/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/958496/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/958496/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/96/958496/3"}}},"commit":{"parents":[{"commit":"2466ef31093b16e8080648744ccf5040bc18e54a","subject":"Hitachi: add support for Hitachi VSP One B20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2466ef31093b16e8080648744ccf5040bc18e54a"}]}],"author":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-25 22:42:24.000000000","tz":-420},"committer":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-26 00:14:00.000000000","tz":-420},"subject":"Hitachi: add support for Hitachi VSP One B20","message":"Hitachi: add support for Hitachi VSP One B20\n\nChange-Id: Ic4b204402fec482d632496f31ef9a2686ffc5d09\nSigned-off-by: Anthony Gamboa \u003canthony.gamboa@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b59053a000c74b22004ae93b75c508da57e95f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b59053a000c74b22004ae93b75c508da57e95f33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
