)]}'
{"id":"openstack%2Fcinder~958569","triplet_id":"openstack%2Fcinder~master~Ib0d8bd57374434003a8844279318778709dd6954","project":"openstack/cinder","branch":"master","topic":"support_safemode","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2026-03-11 14:36:30.000000000","reason":"Change was submitted"},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2026-03-11 14:36:30.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-03-11 14:36:30.000000000","reason":"Change was submitted"},"36962":{"account":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"last_update":"2026-03-11 14:36:30.000000000","reason":"Change was submitted"},"38067":{"account":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"last_update":"2026-03-11 14:36:30.000000000","reason":"Change was submitted"},"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2026-03-11 14:36:30.000000000","reason":"Change was submitted"},"37940":{"account":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"last_update":"2026-03-11 14:36:30.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-03-11 12:58:20.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2026-03-11 14:36:30.000000000","reason":"Change was submitted"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-03-11 14:36:30.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-03-11 14:36:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib0d8bd57374434003a8844279318778709dd6954","subject":"[Pure Storage] Add SafeMode support","status":"MERGED","created":"2025-08-26 18:20:30.000000000","updated":"2026-03-12 00:25:09.000000000","submitted":"2026-03-11 14:36:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958569-support_safemode","meta_rev_id":"fbcda5a99406960748216f3356e5280fcfc9de22","_number":958569,"virtual_id_number":958569,"owner":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},{"value":0,"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-11 14:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-03-10 19:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":2,"date":"2026-03-11 12:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},{"value":0,"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":1,"date":"2026-03-11 12:58:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},{"value":0,"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},{"value":0,"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-26 19:11:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-26 20:09:39.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-26 20:32:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 21:03:28.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-26 21:04:42.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-26 22:06:25.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-08-26 22:41:04.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-26 23:57:12.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-03 16:30:35.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-10 05:31:29.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-11 14:05:39.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-09-18 17:59:19.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-26 21:17:57.000000000","updated_by":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"reviewer":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"state":"CC"},{"updated":"2025-10-08 12:43:51.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"CC"},{"updated":"2025-10-09 21:43:39.000000000","updated_by":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"reviewer":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"state":"REVIEWER"},{"updated":"2025-10-28 23:01:39.000000000","updated_by":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"reviewer":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"state":"REVIEWER"},{"updated":"2025-11-21 14:05:19.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-11-21 15:20:00.000000000","updated_by":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"reviewer":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"state":"REVIEWER"},{"updated":"2025-11-24 09:33:06.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"REVIEWER"},{"updated":"2025-12-19 10:22:57.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2026-01-02 16:37:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2026-02-20 15:15:42.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2026-02-20 15:28:09.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-02-25 12:46:18.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-03-07 04:11:44.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-03-07 04:29:55.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-03-07 04:34:30.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-03-07 04:37:55.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-03-08 07:15:23.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-03-10 19:11:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-03-12 00:25:09.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"bd1d7531d2921358243cede7781986531fa43e11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-08-26 18:20:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebcd96da23096c8bb8096573dfc3510fb9891e2a","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-08-26 18:20:47.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"86794ec21c11ee086e83d5cbccd536331a8e9353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-08-26 18:34:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b8aadbf76142fbdf57639a7d84d923b1f6161c7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 19:11:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41519a3c7d174991a7dcf86821d99f80\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b96d0c963beb470aaeac67eecacf1c92 : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2b3a297738c745f69b235bbb32345c84 : SUCCESS in 19m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7dda90b8198a4768b2ec506490e0eedb : SUCCESS in 30m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b05065508f88fe5098d6abd97372e1dd9b912737","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-26 20:09:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9f48e61ec38c49878e9ce56982ff7224 : SUCCESS in 48m 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/76f80d447244412e85aa5a236b872e8f : FAILURE in 1h 12m 02s","accounts_in_message":[],"_revision_number":2},{"id":"85971abd58e11d1df5efb0492ba63ccb21762c09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 20:32:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41c2cf22b12a480084abc03447e7fbf5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/714343020f7b4cd7a3f1ec2671e8f7c9 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22aa293078b840dd8f185df01ab4b839 : SUCCESS in 11m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a474e7249d7416093ec65976d3c44c8 : SUCCESS in 12m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/287d06e737504f8d9b74fc360151366e : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c3bc979d3b5420b91afcea5e8c05f90 : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/05a0cfc763f5448fa884726123376651 : SUCCESS in 34m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/036ab1ddbc2f43bb824a3afb2ce3493b : SUCCESS in 39m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1fd29f9c836442e38db021ffded022cd : SUCCESS in 1h 19m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87fe54b0d1304f0083938ae620ff4834 : SUCCESS in 32m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06bf7f16077c4207bb41aca2871b7118 : SUCCESS in 13m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/897f4157da14415aa0472548d2e6241e : SUCCESS in 16m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bfb1c97fbc984e0a9bcf1ae6695322d9 : SUCCESS in 3m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7db494113f94a519d2f8324e2f332f3 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7549b21685ea47fdb37f5f430317b055 : SUCCESS in 5m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/290746e3df474273a2853cdccc5aa2c8 : SUCCESS in 10m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3a3e5f6d1616444da97b8fccc524cb42 : FAILURE in 33m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5bc781431ee843d3a346eb29d34dbf0b : SUCCESS in 7m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b5a0427269c74556ae4effa7a5f20f3b : SUCCESS in 1h 54m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5e469ef0f14d45de8892700a9fde281e : FAILURE in 1h 34m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2956a50de4c347f7a3178d2faac06c51 : SUCCESS in 1h 44m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cf25054c4dba4362bf8473ce448eb5d2 : FAILURE in 17m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/29bb4b96fe9744b6b89ea288c265cbfe : SUCCESS in 34m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9214893a306b46f286e241fb393081b0 : SUCCESS in 1h 21m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/84d1ed8d079540979aa1380df9c58d82 : SUCCESS in 1h 38m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/33ff32656c0c497181f215c6c2e02078 : SUCCESS in 34m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c37b1ede8c0c4ff0af6b44abe9cf3c8c : SUCCESS in 1h 29m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/59e7f90373e84be1abd496d6cab58e03 : FAILURE in 21m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d14af4fe603b4680a683fb5000b7d1e3 : SUCCESS in 55m 46s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1453b6e3372d4f769c5e4e505d1cfa7f : SUCCESS in 1h 22m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b13aa026881490eb1ea77e47c754f16 : SUCCESS in 41m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9c7aa07ec8adb211cc8b9ba9a22b373ddbc4d9ac","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 21:03:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1574/.","accounts_in_message":[],"_revision_number":2},{"id":"f62e0e6af7bc16efec53b890fb5450d086533a70","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 21:04:42.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/313/","accounts_in_message":[],"_revision_number":1},{"id":"46a9020df9a85e11070c45238f693aec2c01aa66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-08-26 21:11:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"de1c58e649f46f7b5f9d9666d2f4d05336b09ad8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 22:04:20.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"060a500bb54bcf1b37f16e94b0dff57bdac5071f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 22:04:22.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/314/","accounts_in_message":[],"_revision_number":2},{"id":"ced363536f85da92a2a9196ce11cfe58b8f35910","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-26 22:06:25.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-1/FC/ : UNSTABLE in 3m 56s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-1/iSCSI/ : FAILURE in 3h 44m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"8a48f5db3dac10ba8e546a20bdea503dc18eb466","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 22:12:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57877f4e5bdd4281a45acf7ac9d17713\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/326b8b65f93d4202825dc3171d10e808 : FAILURE in 28m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b066e36b457548bf8eba34c45342ad36 : FAILURE in 32m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/895d71dcccc440b59e4f71b6c82415b2 : SUCCESS in 24m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4a95a4eb7768c99f936cb3848b2a1d3d529f532","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-26 22:41:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958569 : FAILURE in 1h 24m 51s","accounts_in_message":[],"_revision_number":3},{"id":"23acb9b8e7af207edb93745b28205bc77f5fd4f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 23:00:43.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"d5b3b964b260ad711e8cee14c1dce9d1e73e183b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 23:09:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1576/.","accounts_in_message":[],"_revision_number":3},{"id":"245c09766651f3f8243271537c04e58b77e9838c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 23:21:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1711/.","accounts_in_message":[],"_revision_number":3},{"id":"18c86f99de0f7a1a78868bd59bce8d74500fdd8c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-26 23:46:35.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9c9a2cb198db4025aff56276a111b479 : SUCCESS in 53m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/579232195efa4f469af9696f3eeae3d1 : FAILURE in 57m 36s","accounts_in_message":[],"_revision_number":3},{"id":"1bb3df149264d6bdf39be1ac3b89a316417fb373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 23:55:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3db95a25fe5483188372140632ddc49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0393079a47df45609caf9a128d645b1d : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d773b93bb4bd4afc96b1a534ef39608a : SUCCESS in 11m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/36f4eb3e477e469e96f56838f010a401 : SUCCESS in 8m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ff8853155a8435c9db36bf29f0e0d2c : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f6165df57e44e88b6adcd923bd33d4 : SUCCESS in 14m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/f84dcaec78254bd2a297f3a5cb775da8 : SUCCESS in 45m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df79db7a3a4d4a45bd7fd5f523901dd7 : SUCCESS in 55m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0307acea4da345c8a4e750994d5fd055 : SUCCESS in 1h 33m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fd63f6d985d4495a71d0120be6d13fd : SUCCESS in 1h 03m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/817a12048c7d40f6bf23ea916c444b93 : SUCCESS in 18m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce09aca0baaf43f09c67c37c888933a0 : SUCCESS in 21m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aceb729e698844a7a23625943f83f817 : SUCCESS in 5m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d327220cefe9475caeabfff761838e38 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/960a75c95f1342169c2c9a4a5d4acfa9 : SUCCESS in 5m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c8bb54b4371244eda6800d4c77e516c9 : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/82d458e181dd46ed865eb53922671be2 : FAILURE in 36m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/22e10661554348a69109189541ee5a1c : SUCCESS in 4m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/29f76ff708e940d7b63c2b28d4894f75 : SUCCESS in 1h 24m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6e4b9dbdec814663a3a09d7f00f4ec83 : FAILURE in 2h 38m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/827d89d62cca42bc9ded694f256dcc85 : SUCCESS in 1h 45m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/591f881b67c34bcfba0027c070f44884 : FAILURE in 20m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e55a6ecc3c294159a893ba38ad67df8b : SUCCESS in 34m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/674385724ee54026bb13abbc02773377 : SUCCESS in 38m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2d7b65400c4b4b2bae58f839f9f5acd7 : SUCCESS in 1h 37m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4d39fa8743b0474fbcf1e9d2f66c8355 : SUCCESS in 48m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a6a1506ab6964df9bc33c95b1191f669 : POST_FAILURE in 2h 09m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9149d44697544aeeb82311c3770d6947 : FAILURE in 20m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/523a72467ed7424fa03b27f2df624c9a : SUCCESS in 1h 37m 16s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df495944f7ee4db9ac51eb1e59047a2a : SUCCESS in 1h 29m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17d2eebabb574f9c8baf156cacb79128 : SUCCESS in 1h 03m 44s","accounts_in_message":[],"_revision_number":3},{"id":"6aa110215c36afb659f03a4900f271add0392a3e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 23:56:14.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/316/","accounts_in_message":[],"_revision_number":3},{"id":"7f9e7bf3850e952e30bb05e9edbf66452d349877","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-26 23:57:12.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b9fe05ab8e6d78b68a9d165ca6e89eb75c3ec60a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-27 00:03:27.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"21b4619ae964ab024df0dcbcc66e460cd9bfd5d9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-27 00:30:50.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-2/FC/ : FAILURE in 5h 55m 43s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-2/iSCSI/ : UNSTABLE in 2h 56m 20s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"751b5127d5f3ce228c0bce5dfab36125d6e6f421","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 00:53:07.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"071af38b2adc91ade75658619c19dd15afed5e73","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-08-27 01:28:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9984192a69ba57f13ba715d10ab1992c6c8c4a6a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-27 01:52:31.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"fe9dcea01b79d761aef64b53037a1e5b761a365d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 02:14:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc3fc2df8cba4b949ac7bc11da4f6c52\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f45dca7582644779a47a7e3bafc4caa : SUCCESS in 31m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b08ad1b19234495a9037d9d1b411db0 : SUCCESS in 18m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6cbddb0116ef45b5ae1beb6fb1e5cf3c : SUCCESS in 37m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b8efc422205faec3bb63cf2f2a5e30c80fe9be22","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-27 02:52:18.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-3/FC/ : FAILURE in 5h 20m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-3/iSCSI/ : FAILURE in 4h 7m 27s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"2bd14bf9d7d5e8ac0856f749a4587b449cc9f167","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-27 03:24:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f1a45c054c144c209138551631917109 : SUCCESS in 46m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/aa00c2b0212148b3bc3a42f259171a43 : FAILURE in 1h 02m 22s","accounts_in_message":[],"_revision_number":3},{"id":"a8574823de64116deb84725ecb49312f41dbe174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 04:43:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d57cd6c9d7a743f482e65582933136ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5d37631378c4911b70c012408b1c801 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f56bfd477364447a6b1a82a828195e2 : SUCCESS in 11m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6b81fb60392441c39b1abd4074d5e2af : SUCCESS in 12m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bef9a02bdc49460e8ffbe6c4243e5fa7 : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e49fc96bc147c1a39a18299311d010 : SUCCESS in 13m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/fb671d0db3884647890c0bb771fe6cc6 : SUCCESS in 1h 01m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c9f3fa2414b4430ea396ab376db01b5d : SUCCESS in 56m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b33d86aefde4457a864ae30c53610e5b : TIMED_OUT in 2h 06m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a22f9bcc8f434f0786ed9147e7609aad : SUCCESS in 1h 03m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7a2543fac5d4a0092befdb3a813912c : SUCCESS in 16m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3c3311f33cd24ddfa636c19ab646b92a : TIMED_OUT in 41m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/68cd5b0c216347029912bc048b16e164 : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/05b3ae9d9d974e6f8c5ec0158a13eb69 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/189e2c8fbc6c4b1b81f0fa89ddb3828a : SUCCESS in 6m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5e2fd5bd79fd4149b1884c85d59138ca : SUCCESS in 8m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1a9debbc4f35447a957d71c0b301b4dd : FAILURE in 39m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14219ddec2654fec8be544cf0698ff27 : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/90590fdafab44ad7a0ca727eed0232b3 : SUCCESS in 1h 50m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4a34f74e75ec405db46cc5b4635f36dd : TIMED_OUT in 3h 07m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9a7f9dd15ece437584edef6b3c9da3c9 : SUCCESS in 1h 39m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9c96215818504439905e217cbf9d98f6 : FAILURE in 18m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9fa0ba25ff634011a918ab72cff913f3 : SUCCESS in 34m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/648e03be33014b2b881aaab1e7edc7f2 : SUCCESS in 1h 10m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/91141a4e3b5246ed838a23b17a4d7347 : SUCCESS in 2h 01m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d80b7fe1e62745c78230537d88e05d71 : SUCCESS in 25m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/86da87957d014e72856fafdc5df99617 : SUCCESS in 1h 13m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fd16e5d5bb034771bf6e1071a7a81733 : FAILURE in 18m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b1c945a44ff4690a300f61f70b48c1d : SUCCESS in 1h 57m 57s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8f6aa464145400f9244ddd26714f882 : SUCCESS in 1h 19m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/143a7aacc5594d7995eb3e4f11e64809 : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b4617a81c3271c7ccf5a5a940506bd02183884e0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-09-03 16:30:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/958569/3/20250903191553/NEC-ISCSI : SUCCESS in 5h 44m 21s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/958569/3/20250903191553/NEC-FC : SUCCESS in 6h 14m 39s","accounts_in_message":[],"_revision_number":3},{"id":"1a779bf8b0bfa587c5122312ccb713310c825401","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-09-10 04:31:52.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c741e5dbe6e83b9440ea34a587aa56de617d3e40","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-09-10 04:32:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5d09cbe542ba3c6adbbfeaa5c9da4d89b6caa0d6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 04:32:05.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/519/","accounts_in_message":[],"_revision_number":4},{"id":"786e7f9ec24766c713ef23bbe62eddd406348ea4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-10 04:38:50.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/$JOB_ID .","accounts_in_message":[],"_revision_number":4},{"id":"291287f7bf64b71dbb5fd9fe55470765981ae50e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 04:50:17.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"35cb29a150d3f0309bb5ce87071e416fd8e6bdd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 04:57:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/704bfd5838ee482f9189c6effb838e7b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c1fd18fb54254fbcb2d1edb160bf5393 : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0b995746c7e4fa6bb6157d0c6d81318 : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a175d81ca0744a97ba5a01fdc152ea53 : SUCCESS in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aed24e011e446b27e2b95a122b465b560c9845c2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-10 05:31:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/958569 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":4},{"id":"ef57c8bc1913304bc2463f9eebf99bac411b9562","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-10 05:34:28.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1737/.","accounts_in_message":[],"_revision_number":4},{"id":"5e77870ba3bf44cbc677bd2d6e98d1d1d141434d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-10 05:36:46.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1871/.","accounts_in_message":[],"_revision_number":4},{"id":"b50aa73ae2d184f4b64355666189f80e2f567eef","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-10 05:37:41.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/384646206a024fcf83f1f02c6b7729fe : SUCCESS in 57m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bf307f35bb3743d186c66ec04fdebed7 : FAILURE in 1h 02m 46s","accounts_in_message":[],"_revision_number":4},{"id":"46675e3179ba3483982e3d409735e162ca0928fa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-10 05:48:01.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958569 : FAILURE in 1h 14m 31s","accounts_in_message":[],"_revision_number":4},{"id":"70d6fb1ca4d9beb8df94e943761aafefe17cbf23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 06:08:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db4f81d25e5d49979821643c21fecb6a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4891eeac571a45499d6d5b428881dc75 : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd6b90596d8e495ab4218f77723a6e11 : SUCCESS in 11m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55f2211e45004ec1a4e693c81aef9332 : SUCCESS in 7m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a381ea5e998a4dbc8d16ff4db1d3ebe2 : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a557f66da14750ade269b6c1b6b8dd : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/cb75f338b1b244dfb98a8c0c60e0177a : SUCCESS in 57m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/199dec4235394bc8883c6127eaf2fd79 : SUCCESS in 45m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55aedb13d200402baba713942fb00e5c : SUCCESS in 1h 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69f24cd70fc54f0ea48a7327019b059a : SUCCESS in 59m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a330f54671541fd9ec44eceaa5ee0d1 : SUCCESS in 17m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/457394a2852a46099723c343c338c570 : SUCCESS in 21m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/de8175257f094ecbbf31ed7391e35655 : SUCCESS in 3m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ee1328f5d8f343f393fded37af8dfa7c : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fe9d4fcb89fa47728db08df11d97af79 : SUCCESS in 5m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d8031409d3f34288832b37ab914be128 : SUCCESS in 7m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a42407b3f9c54813a6fbec2c97979623 : FAILURE in 36m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c007f8fe2ada4705b59dcdd3b35b09d1 : SUCCESS in 7m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7b0275e12ad44fa28396351b16795220 : SUCCESS in 1h 20m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a7cd90c2ee0f4b9a822e34cb869d527a : FAILURE in 1h 35m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a12ffd85101f4bbeb3b0d4e348e68895 : SUCCESS in 1h 13m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/75ba58bad6e44bf28888120ba99cd6b5 : FAILURE in 11m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/638e6fa795e542e09c921f50a42aa297 : SUCCESS in 25m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a432c1f901184caebc0b7fd3aef5ac2a : SUCCESS in 1h 05m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3e0ba92238244b2797e790c497e6920b : SUCCESS in 1h 32m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c734efa868db424891d56a7eb93546a1 : SUCCESS in 31m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/863acf1e0c714908bd99e5f0fc9202b2 : SUCCESS in 54m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/85578e95d20a4ad59fea2045238790a0 : FAILURE in 20m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c41535fdcabc402c9a08d2cf9432916f : SUCCESS in 1h 10m 10s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e57a5b4d35e847d9bee4027cf380bae0 : SUCCESS in 1h 21m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a5fa85a6502406588703b93f3aa4629 : SUCCESS in 55m 17s","accounts_in_message":[],"_revision_number":4},{"id":"8ce0fff29d0928af3ec0347437a2a82592696be2","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-09-11 14:05:39.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b506cbfcdce19b21f2930cf267f11e8c91b1a588","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-09-18 17:57:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ae34fb7f61a27aefa80047208b430a797da01ece","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-18 17:58:02.000000000","message":"Patch Set 5:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/626/","accounts_in_message":[],"_revision_number":5},{"id":"4413999cc8fe83779691f019d8ba88c9b5330ba1","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-09-18 17:58:08.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"31dc320473f8a3432c7b770541a4942084b2ceb4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-18 17:59:19.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"a2f14738a24bc204cf84b66ef3e75aad66ac31b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-18 18:16:04.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"5c84ffff66655a0f1552f492dd0cbbca0c31cf8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 18:46:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42caeaf66b3040ffb5c1bd794a5f6a61\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/57d8e584e8dd49e49d17be63af837b13 : FAILURE in 30m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a29e12531b65466790d99afccfa66eb6 : FAILURE in 13m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d01a8a3312bc43f085aaff0aee889902 : FAILURE in 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"848724e7b3491e071d527897a5cb025962502666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-09-18 18:53:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8a22f04983bed633d9db8f9886080f6e42973e1f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-18 18:53:32.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/627/","accounts_in_message":[],"_revision_number":6},{"id":"ef079f606f8cee4c3b04e677fe3a98d997e273b6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-18 18:54:46.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"2017d4575e98d28010a99909a2ff63244454156c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-18 19:03:36.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1837/.","accounts_in_message":[],"_revision_number":5},{"id":"3ba87220d796904ce98d6c000e7b15be58f4f68c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-18 19:05:20.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1971/.","accounts_in_message":[],"_revision_number":5},{"id":"47152faca6c71710acfbd63c23e983b3f1c375cb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-18 19:11:43.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958569/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"a5df3a2dc0a4f62b5e118fbe69fbc023136761c4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-18 19:47:35.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib0d8bd57374434003a8844279318778709dd6954 .","accounts_in_message":[],"_revision_number":5},{"id":"3a0dcf0853fb887ea9caff135a89b9e4be7d3aec","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-18 19:49:53.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5f27ab366e944a9b8ce92eacc1b112ef : SUCCESS in 46m 59s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1e3d44ca71014f5188a6a03603eb8c38 : FAILURE in 53m 41s","accounts_in_message":[],"_revision_number":6},{"id":"646d032c7b72e8bc7850407972af507f40ae16fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 20:06:53.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15a65cfadbeb4fcaa07662398ebdfa87\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4fc74ff46e2344ed859757b31353a2ad : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/38932a18084e46dc84604d151442b2e5 : SUCCESS in 26m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/692daf4ae5244dce804f0273b215d0ee : SUCCESS in 32m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d48d6af83e3f077570e6b31d0c7fe99e658e3ba8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-18 20:06:53.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/958569 : SUCCESS in 56m 23s","accounts_in_message":[],"_revision_number":6},{"id":"72f9debd26358f6bf69a9324a10a95838b638f4f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-18 20:08:02.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958569 : FAILURE in 1h 10m 27s","accounts_in_message":[],"_revision_number":6},{"id":"d78c9db2371cadf3cac073d4caff2e099a5c22fd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-18 20:40:13.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ib0d8bd57374434003a8844279318778709dd6954 .","accounts_in_message":[],"_revision_number":6},{"id":"69d16504e8438aae627dd248b180e43f4370689a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 21:38:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a3470e0b10485e96634bb01789da18\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408bb146a741451dbf85907ddc0198cf : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d7a29482fcd4543bed375fa226abd76 : SUCCESS in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a0277f0ea2dd4026be477497525c74d4 : SUCCESS in 10m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04d933987fad46f78f1fb7f374980b13 : SUCCESS in 15m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3475eecabbd4321ab43b26b36e6928e : SUCCESS in 13m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/71660f6c0a9440de8f1011e054e9711f : SUCCESS in 57m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/957c1430166846c0b9d41e0924b3bb2f : SUCCESS in 1h 04m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d1b80bd61e94ab1b795873e599b50d3 : SUCCESS in 1h 29m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1847d0f746ed436fac9b6407db844541 : SUCCESS in 1h 01m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f35bc4b593b141ccb8908bce2497a813 : SUCCESS in 15m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f18fc8d9795a46ea8c2e92b70281fb42 : FAILURE in 24m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/91b65f106d3142cf90be9c482391a302 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/418043f04f3e4e95b9c26c2d4802f2c2 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b7fb0adedae546d9a5b3b096f8d310a2 : SUCCESS in 5m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e4da83e9a70e4a91b905981f94e681c7 : SUCCESS in 8m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e8ca21cbc903408ebb072e5ddb36d12c : FAILURE in 36m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14a6d7a280814067b78cefcc6aac4722 : SUCCESS in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1b81624744df46c797c890547df9be5c : SUCCESS in 2h 35m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f94b911eb77548899403d86a0af6c970 : FAILURE in 2h 28m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7aa28e72bf44442e87675e56d3733ddb : SUCCESS in 2h 07m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/57a9904fef5f453fb6a846b7e78498f8 : FAILURE in 16m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0edf679e0e854731b748106bc6addbd6 : SUCCESS in 37m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8f4efc3fe6f24f28b446084fe0fcce6c : SUCCESS in 1h 12m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3512870e703541bf833ed6e9e8441178 : SUCCESS in 1h 09m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2a9e404db48944adb0db46cb64e7153c : SUCCESS in 37m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7166cba333654f8e8a4aa7e2b4cb9396 : SUCCESS in 1h 24m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ff9bf2d2b4914b5db9fd88942a7b7318 : FAILURE in 12m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9dd7153ce3254735b24ee7b30b4b070b : SUCCESS in 1h 45m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca34f2c4e1db412e899c1d13f5f7009d : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":6},{"id":"bbe8137a70e6a366a621a25eb4133598db83b7d6","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-09-22 12:57:00.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"d069d78e0eca4c835a389bca16374f5642eb2950","author":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"date":"2025-09-26 21:17:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7b68c2bb320441b8ed6757efd9e59d4795d89d13","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2025-09-26 21:33:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5c10e19e0165e958cc1cee6844728775ad6e2893","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2025-10-08 12:43:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ab5b111deacbe0d898d8878126f68131adb713ca","author":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"date":"2025-10-09 21:43:39.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a6d937732c52a8ff3829d5e21ecb24533524798e","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2025-10-28 23:01:39.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"31ec34bb78f7845e4644c5e0333efcd689058b42","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-21 14:05:19.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"91c258d2d20b5d344f97027522d91ab67e89a450","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2025-11-21 15:20:00.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3d4cfbb54982ea0cc0e560f71c4cd2ca9c83097c","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-11-24 09:33:06.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"90f83b04dc70a32960d5e9ec9cf1b1954a0276ea","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-12-19 10:22:57.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"b0ebbcb3679e4c6a1bc078b8dfdfba67e85adebf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-01-02 16:37:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3efc99093d40b4621decd7674666e3cf247ef61c","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2026-02-20 15:15:42.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9acdefc23d744744665509f3722edcb405978a93","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-20 15:28:09.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f9cc5eaa1106495311d554832f28325587a70bd6","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-25 12:46:18.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"41aeb53776343914d75d9cb1ab61c5023dcfc55c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-07 03:58:23.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"94967510901d0e756ca41eb4e23028c51186be1f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-07 04:03:44.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b438bb2d0d66480180d2826fb3de2ca7 : RETRY_LIMIT in 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6245504d05f3477894e33bf7d0bf00dc : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":7},{"id":"91a9ef6dcde2b666e6b08c5b760a80a015dfb066","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-07 04:11:44.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/0e32f3cf2d48485ba89ec86eef9f0d31 : FAILURE in 12m 55s","accounts_in_message":[],"_revision_number":7},{"id":"b24e8c822f025e5501987a30c0a91cd74abcdebc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 04:26:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/075d61d6c4eb4594ac9fb9499f3388f8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d48232ce937349a9ba70bea053f6795b : SUCCESS in 25m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b89c340df0104f6db260339ac44941ec : SUCCESS in 21m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/212cef061905423ebbf0ad5be76f6fd6 : SUCCESS in 18m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8b3d4ee9e46385e3a5621cbde26e1281986ad5e1","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-07 04:29:55.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver build/55dfaef357e84454938be6cd81a3bc7a : FAILURE in 31m 02s","accounts_in_message":[],"_revision_number":7},{"id":"aa2f1619755dc190f4ee9e146d116e5cd81b7634","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-07 04:34:30.000000000","message":"Patch Set 7:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/958569 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"78b8f50fcd4b046f4a13ed9fed28acd0041c835c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-07 04:36:36.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958569 : FAILURE in 36m 53s","accounts_in_message":[],"_revision_number":7},{"id":"71959552b02fef36c16e8f0fc89307e6a85e5bc3","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-07 04:37:55.000000000","message":"Patch Set 7:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/958569 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"1e2950d2e223323f5fce047005cdccf6b0b1c705","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-07 04:38:40.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ib0d8bd57374434003a8844279318778709dd6954 .","accounts_in_message":[],"_revision_number":7},{"id":"24fefd90fd05943dcee18d8ebba6a45dc302c855","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-07 04:43:01.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/d1a2384a75ab405ab16b883fb7880207 : FAILURE in 44m 03s","accounts_in_message":[],"_revision_number":7},{"id":"db34f95d91a0dff4ed3cc48be376ad0bfe35a4b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 06:29:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad6fa9a902f545b4bc8e96c22d454fad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2111106cb9cf4d0b97c1e0f3894eaca5 : SUCCESS in 6m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe770afa73bb445db8d38445dffb96af : SUCCESS in 11m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f688b9886d034775ae6762eabb4b0037 : SUCCESS in 6m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/208610f2d84c49359d66d08f2a0121e5 : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37bcab77ae18452bb93d9c5f791de0a4 : SUCCESS in 14m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/744f4e5709c64273b9974af08c736da3 : SUCCESS in 49m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/68ec58d81b6b44ac92314b2f6714dd31 : SUCCESS in 34m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75899d3c60f042d4b0e4f4757c5fecc9 : SUCCESS in 1h 34m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1d02eaadfc14e7e82a4688b3af0c376 : FAILURE in 1h 07m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ac96595fd1045939d17eb8323150f59 : SUCCESS in 9m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ea44336c81a146a19e607f62b13afb2f : SUCCESS in 18m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c96a6d52d3b4deb9b3baed8ca32eb61 : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/91739c5e063f4822ab4928b8793acbfc : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1166288210084087ae56f59fceef0f28 : SUCCESS in 5m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e6ca6a6b5b084f0689017e673d2c011a : SUCCESS in 3m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1082fb5fc5d74a4898ed288067a6f242 : FAILURE in 39m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7f914445d94944c8abb86d33ac11f958 : FAILURE in 3m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5392aea913e44a07b6f1c182797471e5 : SUCCESS in 1h 20m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bf4a86da9fa745298f12ed231ebd9d02 : FAILURE in 2h 10m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/951ebe89a75d4d9c99e6b6dc87c11af8 : SUCCESS in 1h 51m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6f82ba45e34b47aaa1709e0d0eb4a6b7 : SUCCESS in 1h 49m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/82adba9f8cfe4cd3bad83219c67060dd : SUCCESS in 24m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3279580e312a4d1599b216b5ee0d5179 : SUCCESS in 1h 08m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/25a39ce23081400cb4d40d79e50f6eb8 : SUCCESS in 1h 24m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a043c5991bd34571b6b790c107f1c6c6 : SUCCESS in 26m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/378f7235bab14a43858655ab9d91fc97 : SUCCESS in 1h 48m 54s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/00fde4a59ce440aab5798aa62d0d5e8d : FAILURE in 1h 31m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad96497e5ab94a32b5668a52a9059a93 : SUCCESS in 1h 41m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b4c09720052493eae31d9299513b1a4 : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":7},{"id":"0fa6e576dbab746b441c276fbb9e383c0675a840","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-07 06:32:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"eec8f5365e50d1b31b6b4e72127499ed03d6b66a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-07 06:38:25.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/be198e0c6beb434d8c907444a14884ab : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ad3b4eab93604df997f5f7e4fde09ff3 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":7},{"id":"599dc9cb666d9319b4badb0cc550b349c59780e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 06:51:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdee42ee62664b5ea98b62e6443ee8c0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dab9bc5076cd479f83c6d6e7abc4536d : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dcb43e2b1f0d4bdca35b6e2fd3308907 : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0bfffcf7586f4ac2a70f23e01fd4edd7 : SUCCESS in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c24043c02e3d9d7b7d55771db8ed4f3de46f5dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 09:12:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b167a02c2fd48419ff609300fcc8e44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90074cebf9a44aa89a1490b329797a67 : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e82fce473371433db82efdf8ff2d6ea4 : SUCCESS in 11m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0dd2bf2a218943b0ba3738c5f1842ff8 : SUCCESS in 10m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ede717d8207645cca1c03043a071cfa3 : SUCCESS in 11m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb17f6f5bb946208edbc2db31931f32 : SUCCESS in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/6e49ac6f33d84f3ba1e207bb760dbb0c : SUCCESS in 1h 00m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73cf70bf1d1a4db0a4cd6bd63bd5b272 : SUCCESS in 59m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5c0e76e878948c3b84261411c51e29c : SUCCESS in 1h 37m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70da2c75968041a0bb000e3b063b0f75 : SUCCESS in 1h 04m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68ea46f90f564c0fa7120b0917610e27 : SUCCESS in 17m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d258cac55176477195afc51a686ebc5f : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/40e51c30ee7e462fa2be053c2f7e1133 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f74e6fff14884e4ca864c2827c58e90c : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bce677c0e085467caadb3fe84b6d5407 : SUCCESS in 7m 05s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7fe8122a8f4042888077557e50f6f88d : SUCCESS in 9m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/88e9a7c3c33640c3985b1d9d63adede1 : FAILURE in 39m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8952bf2c9524470dbe8c0db77c2ab6ab : FAILURE in 5m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9c68065c45104e10b7d745f771d37bd7 : SUCCESS in 1h 47m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8453c5a26d5f433cb041f8bd76b4200f : FAILURE in 2h 34m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/19cbcc4fd00e41ac90d7e288f907a141 : SUCCESS in 2h 00m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/dd4fcaa78d16493b91e302b7c2129d82 : SUCCESS in 1h 32m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6a690c53dff44706b09733cf33c8583e : SUCCESS in 35m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/040547928bb94470ae9befe4ae375743 : SUCCESS in 1h 13m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/424998561316452295c43b1396e13ced : SUCCESS in 1h 33m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0606c5bf34df473eadc4cbac1dfd9413 : SUCCESS in 25m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0375b66b472b40dab5b74909af3af63b : SUCCESS in 2h 05m 32s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/20b579070e4440499816ede4e6c596c5 : FAILURE in 2h 12m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/593626c07cae4941861618f3a0295f9a : SUCCESS in 1h 43m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04ea66b3bbbe439d9934ce3304c5429b : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":7},{"id":"106197b542c9d69129f62d19e51c584ae3fea336","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-07 18:41:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3210/.","accounts_in_message":[],"_revision_number":7},{"id":"307a1a85ad300a3bed0bd6accb6c18d7f4c81f1f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-07 19:01:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3383/.","accounts_in_message":[],"_revision_number":7},{"id":"04ead36f3db273d0862adbe3c410eafc41479355","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-08 07:15:23.000000000","message":"Patch Set 7:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1008/index.html : SUCCESS in 88m","accounts_in_message":[],"_revision_number":7},{"id":"3fe7e181e8aba843f89eb0e52f02cfc5ca5db9d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-09 22:51:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"3c2480adbdf352d02d4ff7e8987d3b8b8a133456","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-09 22:52:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"58cb6d3cff5cb31c004977d28d8abb931fe79672","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-09 22:57:10.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fbaebf1677fa452889bb4591c30e5d6c : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/af312d87ce0948f799b8bfb5ab875825 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":8},{"id":"dcce2084b71f923c4f3eaf45d274316d0959989d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-09 23:04:44.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/0dc8227838694ffc8f669b73099b5662 : FAILURE in 10m 58s","accounts_in_message":[],"_revision_number":8},{"id":"70c7b78f49a16ebed73bf14888d3bf3804346a1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 23:10:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ad624a1f668435f9e77eca80daf89de\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec7e413155014803a7b57b39f6e3b86c : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f6b3973537a43ebbd6561674dabdf4d : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e03e85dd895a46ba87279236b6fae040 : SUCCESS in 16m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c7441617ccd3651c9e991722898612f7fdef3906","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-09 23:23:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver build/a41f7e47ae3e441084a31794217c6ae3 : FAILURE in 31m 52s","accounts_in_message":[],"_revision_number":8},{"id":"407f5631f94bf499abd5efb568820714f8f50ec9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-09 23:28:43.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib0d8bd57374434003a8844279318778709dd6954 .","accounts_in_message":[],"_revision_number":8},{"id":"0a392bc8283552e92336c6de6bd5a5b151ff4add","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-09 23:31:26.000000000","message":"Patch Set 8:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/958569 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"dfb02053581cbfd158fd0afde81fda3c5ca84f50","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-09 23:31:49.000000000","message":"Patch Set 8:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/958569 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"b080aa8812547001eee823344e121ddfd79e8551","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-09 23:31:58.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958569 : FAILURE in 39m 14s","accounts_in_message":[],"_revision_number":8},{"id":"99c323b577f7f330ced84c2eaf0477f2ef8146eb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-09 23:36:28.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/f12c3ae660054d7daa181cb0431056c5 : FAILURE in 44m 20s","accounts_in_message":[],"_revision_number":8},{"id":"56e9bbadc99ea736636cdbd7ad446a6c292152a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 02:03:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8f1eef3b0d943bbbb3f955c16baccf3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db968c0db81445b4846107156a0a7c90 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3948fa38e2c4eac8c225fa08bcf0004 : SUCCESS in 8m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a878c19719894c9ea8e78d550016894f : SUCCESS in 13m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/59a4494076dc4058a46f1fac4feee9c7 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8fe25e76bdf4366a31ebeb2af019716 : SUCCESS in 13m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/e37f86dcceae42a8b692a4d05fa8e582 : SUCCESS in 1h 01m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c0f14495fe404393b4e7dc53fbc2cffb : SUCCESS in 51m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b3085092da0a447391fc11ee77b85f88 : SUCCESS in 1h 17m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ac5064ded4d421ebf408b62ef226176 : SUCCESS in 1h 02m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04483a723486406199a923df496b35b2 : SUCCESS in 11m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b997ff49992644b990e89ba3bfc7bfb8 : SUCCESS in 23m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ee7f217d5584144a7beb5d7bd8dcf9d : SUCCESS in 5m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b8498e9e633f434f9b5767fb643dd639 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1e7deb2fd96645cbbbfaf65e23baee7e : SUCCESS in 4m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ae27b5917bda48a39faa7800d1eeb6ab : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/10d9a804ada247eb9b5cfebcfb3e8867 : FAILURE in 20m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad3d207920564d78885bbe0ab65b0d2c : FAILURE in 4m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e901d06d5ba44af3ad839e175bfe1732 : SUCCESS in 59m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/819cb71b519048c4a7d0ad0f7e1247be : FAILURE in 2h 17m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/021c15f927284bc98105c62db51bace4 : SUCCESS in 1h 53m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/048e670ab2ec448aab04a22791678018 : SUCCESS in 2h 16m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4a609d9949364f3da814e81bf5d88239 : SUCCESS in 37m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/99438cd3f3dc4b6187838c38a3121397 : SUCCESS in 42m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6385bc9e71541319f654c303e1247d7 : SUCCESS in 1h 44m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/31b0a62a1d804a4dadd5bbdfb3a64584 : SUCCESS in 35m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/db9c08ada26545eeaff615d42a287952 : TIMED_OUT in 3h 07m 28s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d308af06af184fab8cf029d1d0022862 : TIMED_OUT in 3h 07m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7bb3250a0fa0476e83e4c16c06fa29b8 : SUCCESS in 1h 56m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81c555ee35d94ef2832d6acf7c0319b8 : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":8},{"id":"8edd3ad95aeb0da61395c54d9ab72f12795a8829","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-10 04:00:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"72549b4d8fa7f61c3ca52f1eceeabaca6daa94ed","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-10 04:06:23.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d0049e3cce634aaebbd95aec4fad76ad : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8a04cf2197fe40b0a3b65e1bcf53fac5 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":8},{"id":"39c3db47c7fc0a5540833983421ece0839227f8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 04:19:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/519ab8e8bcb747a6985768eedeb8170d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36b6e31b395e479688fce71f68877cdc : SUCCESS in 14m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/048041dc5c6d4fe8a91fc1b558767199 : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2cf8e4b0a03845b4a975bbac1f0bd063 : SUCCESS in 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5518dc56b2f0f502b25fa6df09a49b8f7ca20ebe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-10 04:51:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3408/.","accounts_in_message":[],"_revision_number":8},{"id":"bf33f9383cafb6b913f1632fa519e6e52818cc9a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-10 05:19:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3235/.","accounts_in_message":[],"_revision_number":8},{"id":"0dc0b1a34b19c99ab8b67c0deb1f834ff197425b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 06:13:45.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb6e4188e90447baa55d819c5350bbad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8d6f3bf4d65449a850bfd8bde28744e : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4729ce511d31482b9cdb768cb9ed8d74 : SUCCESS in 11m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa1725bd5e90435190a0a412b7be64f8 : SUCCESS in 6m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d3102cf04c8e45559455592a01aff993 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06120572a9f84185933e2c9cdf6d6c14 : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/fffd0eab912d451cbea2386ccab4b3a9 : SUCCESS in 33m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21c7de39bed6488a82c913f546f84706 : SUCCESS in 30m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/299c7b69ad184f7f93de62dc1048f967 : SUCCESS in 1h 44m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88793dd827f24581b54f035afd61ecab : SUCCESS in 50m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bf56ae8eaba4ff68f21c6bd2f333511 : SUCCESS in 13m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ddde7a1a2ca446059bce36db1a7c4dcd : SUCCESS in 20m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c3e78e2ade3044bfb8527770f6ea1fae : SUCCESS in 6m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/94aa59326ab040a3aa4ac175e9c31d91 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/681a7c3bb9ec4e6791287c86de6f2289 : SUCCESS in 5m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d595196cb5844eeebbe12ca1d9ea4686 : FAILURE in 3m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/067728016fca452f9b74680d0dbe325f : FAILURE in 40m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/27fb3d45e363449488460453eb278564 : FAILURE in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b8cc0dee7f7f4555b570f2729ec32779 : SUCCESS in 1h 49m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/00f40829e0c9401c810261e4a9716e91 : FAILURE in 1h 25m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/89bff895f5924e31a9c3ccb1861d564a : SUCCESS in 1h 46m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d889b8b07c78432081a42214e22636a1 : SUCCESS in 1h 49m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9ef792e01491451f97f2ed119f8ab6fd : SUCCESS in 23m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/025a007a20454b9092eef09e84e5f562 : SUCCESS in 1h 08m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61abe4951c2f47708b66d356d214e465 : SUCCESS in 1h 26m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b12070d9aa41456c9a6ba717df6feba0 : SUCCESS in 43m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7ad6a2c2fa0943978c6a9723cdee1801 : SUCCESS in 1h 44m 25s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d2f63663ff644a46b63b5366a988374b : FAILURE in 2h 08m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7db1f5403e9048918159027c2a91b2cc : SUCCESS in 1h 43m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e28faaaa2198497089e199fa907a2443 : SUCCESS in 34m 23s","accounts_in_message":[],"_revision_number":8},{"id":"7144225c0aa196844449708619de986f54310a2b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-10 08:58:45.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-8/FC/ : UNSTABLE in 5h 8m 58s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-8/iSCSI/ : FAILURE in 7h 32m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"a9a36e7cdd0df5cb50e3dae05f1194c38e938f27","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-10 11:12:09.000000000","message":"Patch Set 8:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1034/index.html : SUCCESS in 92m","accounts_in_message":[],"_revision_number":8},{"id":"55eb248fa748b8055c0ede69fdf705408e42f259","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-10 11:36:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5ad3df13e16182c3523be55945e5f003a8bac743","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-10 11:41:37.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4d7491459596415f8a07d5cbfcc8545e : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6f3455ded8604a84b5d9f0afb5f8cf26 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":8},{"id":"0f5170315a2a45ad9d16926866b7926557a4d835","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 11:55:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73eb198979964cb299e1d7e845c37ec3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6fbada4cc82c4921bda05f242d9ff3be : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c1c7890be30247fba80c4a97ce35b503 : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4bb44af3bc6b4972928db3d441762874 : SUCCESS in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f29093eea9367692eea33439bfd9a96e693c073a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 14:09:06.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d56dab3243ff4719bde80c370fed104e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/516669bd1da2451bb5c5d9cb2741b4ce : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fbefcd55daf4c689179ee54ff71ac4c : SUCCESS in 8m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/22b5fe5c0bc347bb8e79f81748e0ba24 : SUCCESS in 10m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/335f4be1c92943568ef4cb5b8bb774a5 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/550950fd34ea4d6c82ed8d9299339720 : SUCCESS in 13m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/516a7ad76a424da090b8bb14a584cb2e : SUCCESS in 45m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21b82606f7fd4e269664464d2fab1aa6 : SUCCESS in 35m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf64f118f36e4027b7caafcb5b8a3b22 : SUCCESS in 1h 40m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1c1bf91def949d5becabc20fc3aea74 : SUCCESS in 47m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52aa2c489fdc465dbd09ee279ae5e0b7 : SUCCESS in 11m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/54588a98c1354a888909dd064a5b8891 : TIMED_OUT in 41m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c6f40099e5904efebe1d9d518dc1ac94 : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/481465566cb7421ebd73b50c23db7ff3 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b081e9db69142278b32f5e07214b3a4 : SUCCESS in 6m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/663066aa411f49c189a61c5f59129dcf : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6adc584beb7846f99e2da676d13948f1 : FAILURE in 43m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d2ea67a2b34644b7b37119f2419b4d6c : FAILURE in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd18451b85664d35b527e9889f492e5b : SUCCESS in 1h 50m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eeb92e3b28484b9d8c21c5b3b3876c71 : FAILURE in 2h 19m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a9024b5b89a24cc590a2fe67ced22628 : SUCCESS in 1h 52m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e3ce70057a0a4eaca345ed961e9d123b : SUCCESS in 1h 53m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3be6e6bf897a44b49f7804d7cc30e34d : FAILURE in 19m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c6485a8d7ad46858dd42c71702e135f : SUCCESS in 42m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/180143e7b80042a18c2c98976c5bf0e1 : FAILURE in 1h 44m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bb2f2944f57542bfaeed62f2fec67957 : SUCCESS in 31m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/72c2cefd1dd84fbf926c597f01f73764 : SUCCESS in 2h 07m 31s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9c283acba858402286563f34174b587a : SUCCESS in 2h 04m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a691f42f58d04ae4983f1a06e2e9a016 : FAILURE in 19m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b356fd4dd3a54aab9ac89b757635dcdf : SUCCESS in 35m 15s","accounts_in_message":[],"_revision_number":8},{"id":"af79223791ecbc135a5c0fb94d09be992875722f","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-10 14:59:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c95689327a24029b5c6787a593ed3e27ec711ba5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-10 15:09:00.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b45e59570c1c43a8b7a4807f6a2c8152 : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ea1c5a00a2cc4cd78644ca5028c7987c : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":8},{"id":"a0ec271474859860eeb8889ac5b2caa58abb7795","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 15:16:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebdfbf009264422f9e26cdd8dc442c8e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4692eebd8e9426c93c9b61545d3cdef : SUCCESS in 11m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a3630cbef8634ef39a95839903e4f9f3 : SUCCESS in 14m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f16018fea38c44febece3f96153143cf : SUCCESS in 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"074f140f1a4bd3611f7a1535c59914062f8fcbc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 18:08:50.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99bd7593d3cf4a4fbbd909a9c8e56c86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473fcefa5b39499494f3f00c5a162a20 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d518c8091994419aaca1589a4be3e496 : SUCCESS in 12m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab3c5c82f1ac4a14b8de8b95c02d1152 : SUCCESS in 10m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/14e448eb7c994a08991ba286d9b1a6c0 : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80721b54502b48ef89634c5afa145ba8 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/12dd184d31d14a1f8406c03f4054e7f0 : SUCCESS in 1h 04m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2ce94b99383f4ce2b0e63c4746436200 : SUCCESS in 52m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/863c635e12c34ec49e7eb50f13533432 : SUCCESS in 1h 48m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e3caf5dce6740f3b0869a18488c12d2 : SUCCESS in 54m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84c3f61db1384159836216029e244589 : SUCCESS in 16m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8196af3266dc4ff493d63b024cd81d69 : SUCCESS in 26m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7707cbd0bda4da9bf4e2d5a073a3133 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f703dfa2e9e049e68a1c2a0cc30cbadb : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3ba26ffb96e141a48207ae7d316ed03a : SUCCESS in 4m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2f5029ae677541ddb12e4a757fdc15d1 : FAILURE in 3m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/86da34b4bbdc47ed883804858eeadc9f : FAILURE in 51m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1fe60d0f255842ef891ecac7b09f8d27 : FAILURE in 4m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/11a5c8f262b24bab9853441adbff66f0 : SUCCESS in 1h 33m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/20a154beb3d14d7280d77c75118c133f : FAILURE in 2h 21m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/282667dbd7074293b9e4fafec160914c : SUCCESS in 1h 45m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/19519dfef59e475eab79b66714ce6a54 : SUCCESS in 1h 31m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f8be920caa404bee9c531b98df350afd : SUCCESS in 37m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee161b1c12d945ba861a8638d9f6ead4 : SUCCESS in 1h 14m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/23487e7b7bdd4bac826dd4b4a81f42b7 : SUCCESS in 1h 42m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cb9d19025e444870953f2260fa075f19 : SUCCESS in 32m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/08e85350386c4452bb28447a7e1cc894 : TIMED_OUT in 3h 07m 54s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f73ba576a35e40988c261eb675febfd6 : SUCCESS in 1h 04m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3d22670549974f46bd252a4e9e96474b : SUCCESS in 1h 45m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a34ce8c2be354e68ae7c3baf63711311 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":8},{"id":"b3adc08b5d3fb11eaf0c60971510c91ad09cec44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-10 18:26:26.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"190333fdecde8a938efc35c729c4f22870a38488","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-10 18:26:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a89cda5bcad18755c0d7bf1f2b9237d79d20e8ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-10 18:27:16.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"50edb24747ec436d3174ea87b98340a2f3e554a8","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-10 18:27:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"849167deb4f2e936d575fc6e01a41fe7908f04d0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-10 18:33:00.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/17208d20925a4e10b8c4b50c79e810ca : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/011bc6d9bd68421dbf934cdbba294fb5 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":10},{"id":"efdb8709085f370f1399e3dd23ec44eaec4f1669","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-10 18:56:52.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/73b4659ef83a4d89bfae629307f7c7d7 : FAILURE in 11m 34s","accounts_in_message":[],"_revision_number":10},{"id":"9d18ce17fbd0e2294803df55dfbcc7dbe51b0750","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-10 19:03:00.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib0d8bd57374434003a8844279318778709dd6954 .","accounts_in_message":[],"_revision_number":9},{"id":"37aa3a0d24924e1f0a135fdbc1e825d12b17e13c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-10 19:05:11.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib0d8bd57374434003a8844279318778709dd6954 .","accounts_in_message":[],"_revision_number":10},{"id":"37506946dfe91c3319570fd4b8662a772bef5b1e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-03-10 19:11:46.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f065e4d93c5a5b793826ddd3fe61dae63657ec25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 20:42:44.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1c4c8312ffc4f79999d9e21126233c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f2431f2d1342e4aef0f37edaa3987f : SUCCESS in 7m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e43ad3021563426a84c3a377799cb943 : FAILURE in 12m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4df8a15d4dba4642b0222162175a9c05 : SUCCESS in 8m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3bba11fa09c248dfbb3867ba0e27c167 : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef1cdf03fab49ef8550b6bb40aa80d9 : SUCCESS in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/bb1e6362d97e41cab68b25da7efd2951 : SUCCESS in 59m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d043b6c8b7134ea78bac3cb01ffd0b3f : SUCCESS in 59m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4b07996a9564624baad0be69aada283 : SUCCESS in 1h 24m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7055bd199e7c4a2eb4f894c9ee05521e : SUCCESS in 1h 08m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72fb1dfa956b4cdcb197af3431564ab7 : SUCCESS in 11m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/242b3691d0c34f88b6c4c5637459793d : SUCCESS in 29m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2aacf36a6e5b477da6ba8760c07e4069 : SUCCESS in 3m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3fcfb755c7b14f41b1c40f7979330344 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5149443b0ec94b84bb9a645a18047397 : SUCCESS in 4m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/20026a7b5d584114aa1eebda93ba21ba : SUCCESS in 7m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4eb539311b1e4d44aa383bd2aead317c : FAILURE in 28m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f4752da0962463097fcb99b32e3dd14 : FAILURE in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4e7d0cadc82f494f8f21c2c9a035cd5d : SUCCESS in 58m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/04e414e229e14964b1f0f39d5c1987a4 : FAILURE in 1h 27m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/64950219903840b0a04334ca48fe4167 : SUCCESS in 1h 53m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0a85d31910724cea8c26332f59b38a8e : SUCCESS in 1h 55m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/07388e2835844bf2b17f66b4692ac683 : SUCCESS in 39m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bfa144bd68174349b2d23c00a0076c7b : SUCCESS in 1h 12m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fe0221acddb14df487ad8235a788655c : SUCCESS in 1h 36m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fa7a7a1004c3409e9cdf301b274863b4 : SUCCESS in 52m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0a4746d8985549448596fe4e05cd31b2 : SUCCESS in 1h 54m 27s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/c01fed6750b5474cba45c22407f21347 : SUCCESS in 1h 48m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b125947f3734b739d4ff48cbfd3fe68 : SUCCESS in 54m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c20b5231e4884f2285d2540388535ee4 : SUCCESS in 1h 06m 58s","accounts_in_message":[],"_revision_number":10},{"id":"81b67e4dc578ebcabc570fb3f7d6d80b1f29fa3b","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-03-10 20:53:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9b580c007cfd157223ea4900c67267526a0af555","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-10 20:58:31.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/72d124c16aa948f39d62c13dd19dc127 : RETRY_LIMIT in 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a057b1f4cdf8454da846e63a4c03232d : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":10},{"id":"bfa67a7dfa39411a89ae6ff4dbfe0b086eb02f8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 22:20:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f1ec10e0984275b294bb9117eea210\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/652a50d74a8e40ee84c4eadccc7d406d : SUCCESS in 15m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ff0d1f92dea4ea19526d1186ee7fde8 : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6949d54a57ac40c6ba61f64ccf637e9b : SUCCESS in 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"352d9316a8a4e62c833de243806ecf49bfbbe8c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 23:26:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b01793d97e35400c81a4c0f9c4fd55db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9a51486323140a89148d0affb43ed34 : SUCCESS in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71ef255edc4a4e2a8485b1faf96c9417 : SUCCESS in 11m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c85c297fe2474368af7b1089bd140ec5 : SUCCESS in 13m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e05204358352480f971b82f187c90797 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312f4b571d714d7b9b2a18d89c7bfc78 : SUCCESS in 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/de20c890b9f44c58af034866afec8864 : SUCCESS in 1h 05m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29f314521d03487b9331ff7630338cd6 : SUCCESS in 1h 02m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28d5e3a1e79b4cff8580b42461b4e2e9 : SUCCESS in 1h 21m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bea4ae4cbcc448de9be921f9cae2261c : SUCCESS in 1h 07m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0344abc946cc42d889631c0c3f94a08b : SUCCESS in 13m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/82b135b2a6e04aba8fbba01ebadbbfd4 : TIMED_OUT in 41m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d36d38d591084786b3c6cd1e55c3bdb6 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/75d1d19fa1e44ad08eb80294346a908e : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4ac06524313746e8bbf52dd33bcee613 : SUCCESS in 5m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/31f0a9665aa44a5f8f736e2fc6cd2d78 : SUCCESS in 4m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b55b65ea2bd44b34ba62cfe3f7f74854 : FAILURE in 22m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c03092c156c490783458bc03fc8073b : FAILURE in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a42545a6ffaa4db8bb4eae21b8d8cffd : SUCCESS in 1h 58m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f5487ea8db4447eebe1c1c305de76846 : FAILURE in 18m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f3c049a0cc1241beadde82c72cd66f6e : SUCCESS in 2h 12m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/aa919d38768c462b8c7ee3b77a014e21 : SUCCESS in 2h 13m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1cf2863ea138443d977b74cd8bf1fd32 : SUCCESS in 35m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/97f7837735c24d489ab1bd85f2e109c1 : SUCCESS in 43m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/deec44fa1a794ad9b17f5dbb531f8078 : SUCCESS in 1h 31m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5bc14e8ea80a4bbf82354b03ce53ee89 : SUCCESS in 43m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ba67763046f94092b839cb46901e1706 : SUCCESS in 1h 45m 58s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/b32f57b557b64cea97ad4c50e3d09d8b : SUCCESS in 2h 25m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/805f7256210a4fecbdd79ca3b84341a2 : SUCCESS in 2h 06m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29cd728d9f5845ce854d37dc16ebec77 : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":10},{"id":"5287bf2422eeb854260e0f5969f0c9a82ac60801","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-10 23:49:27.000000000","message":"Patch Set 10:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/958569 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"0878cfc71f50cf72eb923c23ba90022a76802201","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-11 00:37:59.000000000","message":"Patch Set 10:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/958569 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"d85dc9cd16cede0e4d8b2ad77b13544ebda378bf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-11 00:50:44.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958569 : FAILURE in 38m 19s","accounts_in_message":[],"_revision_number":10},{"id":"dd71a6add8e5edb15f42464a09f2c5b0e29bfb90","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-11 01:00:08.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver build/3aec5ed3b6b6444d93a6bb0f9ef629fe : FAILURE in 31m 13s","accounts_in_message":[],"_revision_number":10},{"id":"6b3110f394b6cdacad71fd3486c7206886c1b3fc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-11 01:23:26.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2ea6b62bf8f14ff6b8d7f9c1dd0ded41 : FAILURE in 39m 39s","accounts_in_message":[],"_revision_number":10},{"id":"ffb28bfd49f8968eb79d2aa6766f276a5f11a0ff","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-11 04:19:16.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-10/FC/ : FAILURE in 6h 21m 58s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-10/iSCSI/ : FAILURE in 5h 57m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"648ef7c3fd140a2cc2f61b4f989b8a0758990830","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-11 05:49:26.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-9/FC/ : FAILURE in 7h 52m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-958569-9/iSCSI/ : FAILURE in 7h 4m 51s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"d5e1f6adb473522a7c530b5360405af4e467f188","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 08:29:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3259/.","accounts_in_message":[],"_revision_number":9},{"id":"91a372c7d4c155b8ce9a6fef6e415640a6607978","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 09:36:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3260/.","accounts_in_message":[],"_revision_number":10},{"id":"e13444ef6bb52b9dcff17f7168a3526b55f3d8d9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 09:46:29.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3432/.","accounts_in_message":[],"_revision_number":9},{"id":"52efe1e248e372c35ec7f5c802e8b53db80ad773","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 10:54:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3433/.","accounts_in_message":[],"_revision_number":10},{"id":"ffe07b316f3e66e10e4436173dea74db0ba293f9","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-03-11 12:58:20.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"5c099dcf000e091faab2699d553a2e6b22050945","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 12:58:47.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"b1841a2a1c4949dcca34cb177e056ee8034f0b3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 14:36:29.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e61e9ecefee41b6a4d23cf1cf3bd28e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30865bd971814c3c9df18a4dd47cd1e0 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67c67ab0a6b04617baa161d296888bce : SUCCESS in 11m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80884682ffa1422aae62433074f46554 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a08f9e06e354b6b8e773c557d6f0e13 : SUCCESS in 13m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/a997d6a6c4b04c2ba3f8303960079bd6 : SUCCESS in 30m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cd7ab0410c294e52b13c8a3e0b07bbfe : SUCCESS in 58m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/329763c0776241619975b5da830b783f : SUCCESS in 1h 35m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ab770aaee284b1d8c86dda9030caac5 : SUCCESS in 1h 02m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c75cb206c1b4bbe9839b137615a44bf : SUCCESS in 9m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b2086e12b2c74636ac60b20477b7d24b : SUCCESS in 34m 16s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9f33420e2ddf4b67976190753c8c93e9 : SUCCESS in 1h 36m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49b3084aba574d748b8bc090dddf7117 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":10},{"id":"1ce85da6cf222b7d25ed80f50e41e497a02b2289","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 14:36:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"3ed88b8a0d5ec1fa53ca90ec3c13bd5321107be5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 14:38:04.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0567063634df4e9597a3d3d0b69f1036\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/650beba7475f4b5fbc39446944d0878c : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc8d4ee658e64fb7988d88bd0257a571 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":10},{"id":"9960b23062ae8e680c759415c2153f601559e97a","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-12 00:25:06.000000000","message":"Patch Set 10:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1067/index.html : SUCCESS in 98m","accounts_in_message":[],"_revision_number":10},{"id":"fbcda5a99406960748216f3356e5280fcfc9de22","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-03-12 00:25:09.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/1067/ : SUCCESS","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"1a1fe8ab260f84121796298289203dda20843af5","revisions":{"bd27ce3770bfab0bc2c62a3ca3be19ffb7e7f470":{"kind":"REWORK","_number":1,"created":"2025-08-26 18:20:30.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/69/958569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/958569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/958569/1"}}},"commit":{"parents":[{"commit":"7ee3678b162981fa170df5bb12ac81ebdabb37ef","subject":"Merge \"NetApp - Extended Consistency group support for NVMe/TCP driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ee3678b162981fa170df5bb12ac81ebdabb37ef"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:16:46.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:20:20.000000000","tz":-240},"subject":"[Pure Storage] Add SafeMode support","message":"[Pure Storage] Add SafeMode support\n\nAdded a new configuration parameter to enable SafeMode ransomware\nprotection to be applied to OpenStack managed volumes.\n\nPreviously all OpenStack managed volumes were disabled from using\nSafeMode protection and by default, SafeMode protection is\ndisabled.\n\nEnabling this feature may, in rare edge-case scenarios, cause\nunexpected results which shouldb reported to Pure Storage Support\nfor investigation.\n\nChange-Id: Ib0d8bd57374434003a8844279318778709dd6954\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd27ce3770bfab0bc2c62a3ca3be19ffb7e7f470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd27ce3770bfab0bc2c62a3ca3be19ffb7e7f470"}]},"branch":"refs/heads/master"},"09846c76fab34f79ec60b341a86be6f8fbf178f5":{"kind":"REWORK","_number":2,"created":"2025-08-26 18:34:10.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/69/958569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/958569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/958569/2"}}},"commit":{"parents":[{"commit":"7ee3678b162981fa170df5bb12ac81ebdabb37ef","subject":"Merge \"NetApp - Extended Consistency group support for NVMe/TCP driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ee3678b162981fa170df5bb12ac81ebdabb37ef"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:16:46.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:33:54.000000000","tz":-240},"subject":"[Pure Storage] Add SafeMode support","message":"[Pure Storage] Add SafeMode support\n\nAdded a new configuration parameter to enable SafeMode ransomware\nprotection to be applied to OpenStack managed volumes.\n\nPreviously all OpenStack managed volumes were disabled from using\nSafeMode protection and by default, SafeMode protection is\ndisabled.\n\nEnabling this feature may, in rare edge-case scenarios, cause\nunexpected results which shouldi be reported to Pure Storage\nSupport for investigation.\n\nEnabling this feature will also force the pure_eradicate_on_delete\nparameter to be False, as this is incompatible with SafeMode.\n\nChange-Id: Ib0d8bd57374434003a8844279318778709dd6954\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09846c76fab34f79ec60b341a86be6f8fbf178f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09846c76fab34f79ec60b341a86be6f8fbf178f5"}]},"branch":"refs/heads/master"},"3908e3e17d8115bcbfd190d0caa2d511d96acfcf":{"kind":"REWORK","_number":3,"created":"2025-08-26 21:11:54.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/69/958569/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/958569/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/958569/3"}}},"commit":{"parents":[{"commit":"7ee3678b162981fa170df5bb12ac81ebdabb37ef","subject":"Merge \"NetApp - Extended Consistency group support for NVMe/TCP driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ee3678b162981fa170df5bb12ac81ebdabb37ef"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:16:46.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 21:11:38.000000000","tz":-240},"subject":"[Pure Storage] Add SafeMode support","message":"[Pure Storage] Add SafeMode support\n\nAdded a new configuration parameter to enable SafeMode ransomware\nprotection to be applied to OpenStack managed volumes.\n\nPreviously all OpenStack managed volumes were disabled from using\nSafeMode protection and by default, SafeMode protection is\ndisabled.\n\nEnabling this feature may, in rare edge-case scenarios, cause\nunexpected results which shouldi be reported to Pure Storage\nSupport for investigation.\n\nEnabling this feature will also force the pure_eradicate_on_delete\nparameter to be False, as this is incompatible with SafeMode.\n\nChange-Id: Ib0d8bd57374434003a8844279318778709dd6954\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3908e3e17d8115bcbfd190d0caa2d511d96acfcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3908e3e17d8115bcbfd190d0caa2d511d96acfcf"}]},"branch":"refs/heads/master"},"9a13f086feb56465b40bf1e85250535f8f1ece66":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-09-10 04:31:52.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/69/958569/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/958569/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/958569/4"}}},"commit":{"parents":[{"commit":"dd04417ad559eb3db5bff732a5db50df3a8ca729","subject":"Merge \"Register glance user in keystoneauth plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd04417ad559eb3db5bff732a5db50df3a8ca729"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:16:46.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-09-10 04:31:52.000000000","tz":0},"subject":"[Pure Storage] Add SafeMode support","message":"[Pure Storage] Add SafeMode support\n\nAdded a new configuration parameter to enable SafeMode ransomware\nprotection to be applied to OpenStack managed volumes.\n\nPreviously all OpenStack managed volumes were disabled from using\nSafeMode protection and by default, SafeMode protection is\ndisabled.\n\nEnabling this feature may, in rare edge-case scenarios, cause\nunexpected results which shouldi be reported to Pure Storage\nSupport for investigation.\n\nEnabling this feature will also force the pure_eradicate_on_delete\nparameter to be False, as this is incompatible with SafeMode.\n\nChange-Id: Ib0d8bd57374434003a8844279318778709dd6954\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a13f086feb56465b40bf1e85250535f8f1ece66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a13f086feb56465b40bf1e85250535f8f1ece66"}]},"branch":"refs/heads/master","description":"Rebase"},"43e1ee0108b9745d721eb38c477a6960ff2f80a8":{"kind":"REWORK","_number":5,"created":"2025-09-18 17:57:50.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/69/958569/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/958569/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/958569/5"}}},"commit":{"parents":[{"commit":"569b4679f7fd808bd9aa70cc70d22546a9fdb293","subject":"Merge \"[NetApp] Enabling total_volumes capability support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/569b4679f7fd808bd9aa70cc70d22546a9fdb293"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:16:46.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-09-18 17:57:32.000000000","tz":-240},"subject":"[Pure Storage] Add SafeMode support","message":"[Pure Storage] Add SafeMode support\n\nAdded a new configuration parameter to enable SafeMode ransomware\nprotection to be applied to OpenStack managed volumes.\n\nPreviously all OpenStack managed volumes were disabled from using\nSafeMode protection and by default, SafeMode protection is\ndisabled.\n\nEnabling this feature may, in rare edge-case scenarios, cause\nunexpected results which shouldi be reported to Pure Storage\nSupport for investigation.\n\nEnabling this feature will also force the pure_eradicate_on_delete\nparameter to be False, as this is incompatible with SafeMode.\n\nChange-Id: Ib0d8bd57374434003a8844279318778709dd6954\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43e1ee0108b9745d721eb38c477a6960ff2f80a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43e1ee0108b9745d721eb38c477a6960ff2f80a8"}]},"branch":"refs/heads/master"},"2bd22e7ad8c1ca380ff3f0d47af4f2e1b6072428":{"kind":"REWORK","_number":6,"created":"2025-09-18 18:53:21.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/69/958569/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/958569/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/958569/6"}}},"commit":{"parents":[{"commit":"569b4679f7fd808bd9aa70cc70d22546a9fdb293","subject":"Merge \"[NetApp] Enabling total_volumes capability support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/569b4679f7fd808bd9aa70cc70d22546a9fdb293"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:16:46.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-09-18 18:53:10.000000000","tz":-240},"subject":"[Pure Storage] Add SafeMode support","message":"[Pure Storage] Add SafeMode support\n\nAdded a new configuration parameter to enable SafeMode ransomware\nprotection to be applied to OpenStack managed volumes.\n\nPreviously all OpenStack managed volumes were disabled from using\nSafeMode protection and by default, SafeMode protection is\ndisabled.\n\nEnabling this feature may, in rare edge-case scenarios, cause\nunexpected results which shouldi be reported to Pure Storage\nSupport for investigation.\n\nEnabling this feature will also force the pure_eradicate_on_delete\nparameter to be False, as this is incompatible with SafeMode.\n\nChange-Id: Ib0d8bd57374434003a8844279318778709dd6954\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bd22e7ad8c1ca380ff3f0d47af4f2e1b6072428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bd22e7ad8c1ca380ff3f0d47af4f2e1b6072428"}]},"branch":"refs/heads/master"},"10982b92c83d984ab28e54420aa53ad657e8d182":{"kind":"REWORK","_number":7,"created":"2026-03-07 03:58:23.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/69/958569/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/958569/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/958569/7"}}},"commit":{"parents":[{"commit":"df837708808e0580039d88e73fdaa3ebcc86ea0c","subject":"Update docs for Pure rebrand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df837708808e0580039d88e73fdaa3ebcc86ea0c"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:16:46.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-03-07 03:57:46.000000000","tz":-300},"subject":"[Pure Storage] Add SafeMode support","message":"[Pure Storage] Add SafeMode support\n\nAdded a new configuration parameter to enable SafeMode ransomware\nprotection to be applied to OpenStack managed volumes.\n\nPreviously all OpenStack managed volumes were disabled from using\nSafeMode protection and by default, SafeMode protection is\ndisabled.\n\nEnabling this feature may, in rare edge-case scenarios, cause\nunexpected results which shouldi be reported to Pure Storage\nSupport for investigation.\n\nEnabling this feature will also force the pure_eradicate_on_delete\nparameter to be False, as this is incompatible with SafeMode.\n\nChange-Id: Ib0d8bd57374434003a8844279318778709dd6954\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/10982b92c83d984ab28e54420aa53ad657e8d182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/10982b92c83d984ab28e54420aa53ad657e8d182"}]},"branch":"refs/heads/master"},"64a0e69f35d534db7f3272e2da1ad707732f215d":{"kind":"REWORK","_number":8,"created":"2026-03-09 22:51:34.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/69/958569/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/958569/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/958569/8"}}},"commit":{"parents":[{"commit":"df837708808e0580039d88e73fdaa3ebcc86ea0c","subject":"Update docs for Pure rebrand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df837708808e0580039d88e73fdaa3ebcc86ea0c"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:16:46.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-03-09 22:51:16.000000000","tz":-240},"subject":"[Pure Storage] Add SafeMode support","message":"[Pure Storage] Add SafeMode support\n\nAdded a new configuration parameter to enable SafeMode ransomware\nprotection to be applied to OpenStack managed volumes.\n\nPreviously all OpenStack managed volumes were disabled from using\nSafeMode protection and by default, SafeMode protection is\ndisabled.\n\nEnabling this feature may, in rare edge-case scenarios, cause\nunexpected results which shouldi be reported to Pure Storage\nSupport for investigation.\n\nEnabling this feature will also force the pure_eradicate_on_delete\nparameter to be False, as this is incompatible with SafeMode.\n\nChange-Id: Ib0d8bd57374434003a8844279318778709dd6954\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64a0e69f35d534db7f3272e2da1ad707732f215d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64a0e69f35d534db7f3272e2da1ad707732f215d"}]},"branch":"refs/heads/master"},"c4e61b6c6d79ed57927e71d65cb43810136ecf4e":{"kind":"REWORK","_number":9,"created":"2026-03-10 18:26:26.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/69/958569/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/958569/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/958569/9"}}},"commit":{"parents":[{"commit":"df837708808e0580039d88e73fdaa3ebcc86ea0c","subject":"Update docs for Pure rebrand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df837708808e0580039d88e73fdaa3ebcc86ea0c"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:16:46.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-03-10 18:26:15.000000000","tz":-240},"subject":"[Pure Storage] Add SafeMode support","message":"[Pure Storage] Add SafeMode support\n\nAdded a new configuration parameter to enable SafeMode ransomware\nprotection to be applied to OpenStack managed volumes.\n\nPreviously all OpenStack managed volumes were disabled from using\nSafeMode protection and by default, SafeMode protection is\ndisabled.\n\nEnabling this feature may, in rare edge-case scenarios, cause\nunexpected results which shouldi be reported to Pure Storage\nSupport for investigation.\n\nEnabling this feature will also force the pure_eradicate_on_delete\nparameter to be False, as this is incompatible with SafeMode.\n\nChange-Id: Ib0d8bd57374434003a8844279318778709dd6954\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c4e61b6c6d79ed57927e71d65cb43810136ecf4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c4e61b6c6d79ed57927e71d65cb43810136ecf4e"}]},"branch":"refs/heads/master"},"1a1fe8ab260f84121796298289203dda20843af5":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-03-10 18:27:16.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/69/958569/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/958569/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/958569/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/958569/10"}}},"commit":{"parents":[{"commit":"df837708808e0580039d88e73fdaa3ebcc86ea0c","subject":"Update docs for Pure rebrand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df837708808e0580039d88e73fdaa3ebcc86ea0c"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2025-08-26 18:16:46.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-03-10 18:27:16.000000000","tz":0},"subject":"[Pure Storage] Add SafeMode support","message":"[Pure Storage] Add SafeMode support\n\nAdded a new configuration parameter to enable SafeMode ransomware\nprotection to be applied to OpenStack managed volumes.\n\nPreviously all OpenStack managed volumes were disabled from using\nSafeMode protection and by default, SafeMode protection is\ndisabled.\n\nEnabling this feature may, in rare edge-case scenarios, cause\nunexpected results which should be reported to Pure Storage\nSupport for investigation.\n\nEnabling this feature will also force the pure_eradicate_on_delete\nparameter to be False, as this is incompatible with SafeMode.\n\nChange-Id: Ib0d8bd57374434003a8844279318778709dd6954\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a1fe8ab260f84121796298289203dda20843af5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a1fe8ab260f84121796298289203dda20843af5"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
