)]}'
{"id":"openstack%2Fcinder~958582","triplet_id":"openstack%2Fcinder~master~I4cb76c8307da2dc59a564843c69f05eacf337dc0","project":"openstack/cinder","branch":"master","topic":"bug/2120737","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-08-27 20:21:15.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4cb76c8307da2dc59a564843c69f05eacf337dc0","subject":"[WIP] use load_session_from_conf_options to create glanceclient","status":"NEW","created":"2025-08-26 20:30:07.000000000","updated":"2025-09-04 07:10:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e4bb97bd9bc5461ca296fbc2ca9b4138c7a1c768","_number":958582,"virtual_id_number":958582,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-27 19:31:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-26 20:31:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-26 20:31:34.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-26 20:31:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 22:00:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-26 22:48:14.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-26 23:00:44.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-27 00:20:46.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-08-27 01:50:19.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-04 07:10:15.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"35914cf64678458e755d9b59797572ac9a1d5ac0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-26 20:30:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43fee4393482f4ad454f018b02e3437557a23e60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 20:31:19.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/958567 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"24a04d8b7e1b5a16c088f50fef024c264ac0dba1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-26 20:31:34.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"5cce5b8e0494cd3d02345def1610e380809db46e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 20:31:47.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/958567 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"7df80781750ffa6928283b2c7424aa651aac35f9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-26 20:33:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31bdadea4343ba020426ee5816bde7ad8ebb1864","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-26 20:38:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0da807b1d9c9f861f59277cfe5eff54bc2ee6534","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 22:00:17.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1575/.","accounts_in_message":[],"_revision_number":1},{"id":"76b3ce80182d01a3b76e4c5dcea4be4d0ca16bc7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-26 22:05:53.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1710/.","accounts_in_message":[],"_revision_number":1},{"id":"6b15a61b855ec1c830d5e4a819f4acc0fdaf291e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 22:19:36.000000000","message":"Patch Set 1:\n\n(34 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98ec8e1f9dac4110aaf5133d21683d8a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83119a5b34284fd09a9c4449779118e9 : FAILURE in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d10f673aff85499e8db8b2c43b4723d8 : SUCCESS in 14m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4468460b33cb45aa977a0c330e4d5f1e : SUCCESS in 12m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51bf774af10146d28ee7dfde66e6d604 : SUCCESS in 17m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2742b6d5031441119ae63e37c89bb433 : SUCCESS in 13m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/19cfa030db304f81ad7e1a097f8006cb : FAILURE in 34m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5449486877c24adb8c429c4710b9b221 : FAILURE in 1h 06m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3c5a139509b47748977f00985b18d2e : FAILURE in 1h 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b97a5f32e3d4c1f8871f4ef47c74b76 : SUCCESS in 55m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c29045a2c16e41d2be553303adbfb924 : SUCCESS in 24m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bf1545e1d2644229b821eb7aaf41c93b : SUCCESS in 6m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6393a631945549e18c23c6962ad5de4f : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d3387da3ebc04513ba39fe96d737a2f5 : SUCCESS in 5m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/75dc8d32aa4243798aa316397843e590 : SUCCESS in 7m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5c25b793c756412a82eb0cfcc18c9e6a : FAILURE in 25m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b7e435548504957a37f102c33e40648 : SUCCESS in 6m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2c0b6d1a9bb94eddaf04ba8e4dcde65b : FAILURE in 1h 25m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6c8972d182be40ef93c477b9d9bdfde1 : FAILURE in 1h 30m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d8de85a59dbb484283854bb4ddf81513 : FAILURE in 1h 40m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/059fff8da73a46d8b787ff1376bff628 : FAILURE in 11m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/729b0f0b2eac426991bc0e0f303ff733 : SUCCESS in 37m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/de59be3d8c7148ca904b38672d08514f : FAILURE in 1h 12m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0ade7ec2de774e88a535daae6908501c : FAILURE in 1h 22m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/967ec3812efa468d82a47ea9bb0fef1d : FAILURE in 30m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fbddc1b5d67d4b07b5d2f181229860c7 : FAILURE in 49m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6c757a315cda449ba31b79eed39d7cb2 : FAILURE in 13m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7098324c16864039ba495a875e0cef2f : FAILURE in 1h 19m 05s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fce5191503904c8dbf56b2faefe2a94d : FAILURE in 1h 09m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bc60b00cf8542388946780832ce67ba : FAILURE in 43m 26s","accounts_in_message":[],"_revision_number":1},{"id":"5d6470d8956675a68c5c7eb6f51890b4970aec7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 22:42:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/171cb6e0897c4288841fc88721eb0b2c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce102825248045bdbd5355281940976e : SUCCESS in 22m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c2661efb349843678e643c4ba9c7daee : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/44075e834cb74a9e85e0d9114397f306 : SUCCESS in 39m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63b61b7453cacf703cd326c2d04130f3c286d68f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-26 22:48:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/96552a8c03d84ea0bfb71dc08f33490b : FAILURE in 44m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b19965946a1d4a1d8972f9861760a60d : FAILURE in 52m 58s","accounts_in_message":[],"_revision_number":1},{"id":"62feeb4f998037893396ab4210eba60b428cf1f6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 23:00:44.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/315/","accounts_in_message":[],"_revision_number":1},{"id":"10303605785a7eeb2625d79ddaac32e962b5ecbc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-26 23:56:13.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958582/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958582/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"24f7d906770d67f4d36507dc10520293c72a5dad","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-27 00:20:46.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-82-958582-1/FC/ : UNSTABLE in 4m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-82-958582-1/iSCSI/ : FAILURE in 3h 49m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"0aa9a56a061876c6ea503b6f7140f1945d40ef7e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-27 01:50:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"61c5fcb65233cd32837d1285fe4a5b50a0de3bf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-27 17:16:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"00033602be174c4b62145702bb480b61ab7cc9e5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 19:26:21.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/327/","accounts_in_message":[],"_revision_number":2},{"id":"9bb4d2995b01e375f9b20ca1019eec907bfac161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 19:31:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e3e218b33ad4f8dbb76b0ebaae060a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae9c399e6e9840a58c597d98dd09b250 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d409a388b0a042309b23f3bcbd0af62c : SUCCESS in 14m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d10b4fc6f6074bdb899271e9453b9cd1 : SUCCESS in 8m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc7fe10a0587467987afbecbcc342d5f : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a4b67cd899846d68ad737b90b52b0b0 : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/660ac1eb1c5f4b4095d67dd3ae84067b : FAILURE in 52m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/985faefc6b6b44c3a8718f1e6b750724 : FAILURE in 48m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf950719e3fa4f7f9cb0508a29f46abc : FAILURE in 1h 16m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9126f6854a4541fbb5bdafd5f64d2cc1 : SUCCESS in 56m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ef44be87a78f4ce295c3c38a20ddc2c3 : TIMED_OUT in 41m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4183a93df35a4a9db042b6313fa5f130 : SUCCESS in 3m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9edf448f88384b8c99f78428727ad43d : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8df9daf545014576a5e8299082e26fb8 : SUCCESS in 4m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2b5e2ab1eec74af2935bc463b8a1d9b7 : SUCCESS in 14m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3ffb813170854c63b0016c02701996af : FAILURE in 19m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8a787c624bcf494d86a2528928ab216f : SUCCESS in 11m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/743623b11d1a4b72867bb748bc50bffc : FAILURE in 1h 34m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e6998465479a410da9b07514a766b2f5 : FAILURE in 2h 08m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2c644f7726eb4189a39e0c2acf004d18 : FAILURE in 1h 39m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d8ca2942051645b3bae98eed1ce09a1f : FAILURE in 19m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a428843c8cd7457b90dc8935e51d98f4 : SUCCESS in 38m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e0c6138ebba948c3b8f26c423ecc1905 : FAILURE in 1h 12m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/690405a3c0084b06a8c50bbcaa3e40be : FAILURE in 1h 16m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/17e897ed1fb94df0949ec38a658105a1 : FAILURE in 40m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/006ab15914cf4017864d1805f0cbff18 : FAILURE in 1h 01m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e891ec16958e4d8b89ee5ff12b80a7c4 : FAILURE in 16m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/56a836935d544eb684ff46fa3918fac8 : FAILURE in 43m 52s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/490d8aa756d449cfa29fd7cf0c343704 : FAILURE in 1h 06m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4729f0b9c7aa43c49e08babd3f7b53eb : FAILURE in 53m 32s","accounts_in_message":[],"_revision_number":2},{"id":"edda233b3fb905360c11926671789c968dd3b9af","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-27 19:38:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2769dcb123d043c5a280455c59556368 : FAILURE in 43m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/566824729551417ea48b87e2343e88b8 : FAILURE in 54m 01s","accounts_in_message":[],"_revision_number":2},{"id":"1b75ae8e7b43a01409da11843c3a6f9dbc10ba08","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 20:21:15.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958582/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958582/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"5208a8c538da891061f098e5a2df801608ef3113","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-27 20:29:52.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"cb07445fade678d6927039af93a5b6f1e4133973","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-27 20:32:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958582 : FAILURE in 1h 26m 03s","accounts_in_message":[],"_revision_number":2},{"id":"f8c40459ee5742f1282575b8823241002946ae19","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-27 22:04:08.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1589/.","accounts_in_message":[],"_revision_number":2},{"id":"874c5b5c2824899adf1938622cdccb4ac27a0519","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-27 22:12:30.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1724/.","accounts_in_message":[],"_revision_number":2},{"id":"a240b0d3911c9fe0cc8f39dc1783ca986a3ff0c7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-27 22:16:38.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-82-958582-2/FC/ : FAILURE in 4h 56m 19s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-82-958582-2/iSCSI/ : FAILURE in 4h 12m 4s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"328522d038b81ac6bdb0dc9cf200b28ba86ed5ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 00:01:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb5e2b9f7096441a91d1588d5820d388\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cd66da510de0404496103a55522d01bd : SUCCESS in 23m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d8b39611026c451db4c54ebbe89e1ec6 : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c2c485a206f240de8049130121a8a416 : SUCCESS in 45m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e4bb97bd9bc5461ca296fbc2ca9b4138c7a1c768","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-09-04 07:10:15.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/82/958582/2/20250904074626/NEC-ISCSI : FAILURE in 8h 10m 50s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/82/958582/2/20250904074626/NEC-FC : FAILURE in 8h 23m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a9485f527e6266de0b4b6d859f4f4988b88e4d74","revisions":{"e7dc831e2206c6009228d8e59f8fef68193ee78e":{"kind":"REWORK","_number":1,"created":"2025-08-26 20:30:07.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/82/958582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/958582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/958582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/958582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/958582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/958582/1"}}},"commit":{"parents":[{"commit":"630afc75b9bce138f6ca7a03a548eb0d6aa6c972","subject":"Merge \"mypy: Bump to 1.16.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/630afc75b9bce138f6ca7a03a548eb0d6aa6c972"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-08-26 20:27:13.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-08-26 20:27:13.000000000","tz":60},"subject":"[WIP] use load_session_from_conf_options to create glanceclient","message":"[WIP] use load_session_from_conf_options to create glanceclient\n\nThis change tests if using a config section with keystone\ncredentials to create teh user token properly allow interacting\nwith glance with the service role.\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/958567\nCloses-Bug: #2120737\nChange-Id: I4cb76c8307da2dc59a564843c69f05eacf337dc0\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e7dc831e2206c6009228d8e59f8fef68193ee78e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e7dc831e2206c6009228d8e59f8fef68193ee78e"}]},"branch":"refs/heads/master"},"a9485f527e6266de0b4b6d859f4f4988b88e4d74":{"kind":"REWORK","_number":2,"created":"2025-08-27 17:16:59.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/82/958582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/958582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/958582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/958582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/958582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/958582/2"}}},"commit":{"parents":[{"commit":"630afc75b9bce138f6ca7a03a548eb0d6aa6c972","subject":"Merge \"mypy: Bump to 1.16.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/630afc75b9bce138f6ca7a03a548eb0d6aa6c972"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-08-26 20:27:13.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-08-27 17:16:52.000000000","tz":60},"subject":"[WIP] use load_session_from_conf_options to create glanceclient","message":"[WIP] use load_session_from_conf_options to create glanceclient\n\nThis change tests if using a config section with keystone\ncredentials to create teh user token properly allow interacting\nwith glance with the service role.\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/958567\nCloses-Bug: #2120737\nChange-Id: I4cb76c8307da2dc59a564843c69f05eacf337dc0\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a9485f527e6266de0b4b6d859f4f4988b88e4d74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a9485f527e6266de0b4b6d859f4f4988b88e4d74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
