)]}'
{"id":"openstack%2Fcinder~958611","triplet_id":"openstack%2Fcinder~master~I748713c1ac1faece0865ebcd9a273a94d6d0b625","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"37923":{"account":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"last_update":"2025-09-22 23:43:36.000000000","reason":"Change was abandoned"},"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-08-29 01:16:59.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I748713c1ac1faece0865ebcd9a273a94d6d0b625","subject":"Hitachi: add support for Hitachi VSP One B20","status":"ABANDONED","created":"2025-08-27 01:02:18.000000000","updated":"2025-09-22 23:43:36.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2225d04cb29ba9054ee9880759f206b6528401c3","_number":958611,"virtual_id_number":958611,"owner":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-28 23:20:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-27 01:34:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-27 01:45:02.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-27 02:21:18.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-27 03:22:31.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-27 03:35:10.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-27 03:48:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-27 05:21:18.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-27 06:26:30.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-08-28 20:09:17.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-01 14:42:14.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-09-04 20:41:29.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"4c4d247931aacfcae4354674f51b0da4e4a1a06d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-27 01:02:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df14e49635634b8bfe0c2b4abddae8cdf218d894","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 01:34:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e82206ae2d274dfd8e034c2347434cbd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1856ec3239f446fb7c18f1b3817bd62 : SUCCESS in 21m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d29ef8f7646244c29b77e7ef7a1b1e97 : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1a3c22ec136546c8b5a2e05dfcd8413b : SUCCESS in 29m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdcb9b54a4e485484aa8182df2040bd91b171af5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-27 01:45:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958611 : FAILURE in 36m 34s","accounts_in_message":[],"_revision_number":1},{"id":"19a7903142117cb559ecba82b2226022504aab59","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-27 02:21:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/06bf5db69d3b4023856f4b078e8489dc : SUCCESS in 47m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/481c01bb633c4a4d8d3e4422ea3d3b5c : FAILURE in 51m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3fd4a3c2dcdf8420241aca542d0aeb3cc16773a9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-27 03:22:31.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1580/.","accounts_in_message":[],"_revision_number":1},{"id":"f6f40180ecc3c07d34b7c24a921abbdc9709900a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 03:35:10.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/320/","accounts_in_message":[],"_revision_number":1},{"id":"df077f8da92d07f1e0220e0de7fec670c53e1a50","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-27 03:40:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1715/.","accounts_in_message":[],"_revision_number":1},{"id":"e11461e7ac46f2b9ce668bdc5478672a1c550b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 03:48:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88366015867d4093873c6d01d9c120e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea15abdd865442238d2a6459bc85ce47 : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6a3e56d94194ac9804b8f0d3b648381 : SUCCESS in 9m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90d73e1854f74d978bd4098e52fcd0c2 : SUCCESS in 8m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4b0bb1803084ccc96e95aca655d5fcc : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8af7e051ff34435a8bffaf9d327404f1 : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/713b3e6fb96e45f5b8da5f2de95297bd : SUCCESS in 29m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1e28375ab01246dab4ba594c62853682 : SUCCESS in 1h 02m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/65fe9432abab47e8b62a16cc35aa5c9f : SUCCESS in 1h 19m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90d258b12a7841119f877069b81e4a0c : SUCCESS in 44m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/efa83ba70f0a47da9db713e6ceaeac4a : SUCCESS in 16m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3788db58f3934be695fc489a208cfc57 : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5b1d9a45f4d8469b83195392f416f750 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fb87b7e0ee084b52ac6b6ccd028f7f15 : SUCCESS in 7m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5da9686901b8454e910e679796f98c72 : SUCCESS in 8m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/32198f2446f842a1aa58a431aa19d00d : FAILURE in 26m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48a14c16af4c4466b3edae307b4b344a : SUCCESS in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f99728c7e0264911933f52590edb3cd6 : SUCCESS in 2h 04m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b5d6e3e1c2934e5897f4aa1faddd5da5 : FAILURE in 2h 11m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0485a62766164021b61818d545abec91 : SUCCESS in 1h 48m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b9c747484f74406c9155b1232bd73693 : FAILURE in 20m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/878cf4d002f340f2bfb66076835c6d7b : SUCCESS in 27m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1ade5ed6dd8a43c88649c2e9e97c8c43 : SUCCESS in 57m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bb0cd29c1d8d46e5bb52db8500678f2a : SUCCESS in 1h 35m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4cf2d31bb75f49d1913d63e4eba84d70 : SUCCESS in 37m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fa8ea26ff06d458cb8f3d5b0515bc1ce : POST_FAILURE in 2h 40m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1384f2223d9c495caa632321be338d93 : FAILURE in 11m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5d7806021e9442f986ec2a3c5d5bbd9d : SUCCESS in 1h 43m 39s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25fe86bffd81412291238aa5a115930e : SUCCESS in 1h 22m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f862ef47d5548a69827d770152268c5 : SUCCESS in 43m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a8a309499bb357e455dd1f87c6bb98e23d79e7b4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 04:28:53.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958611/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958611/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"1b54fcf909fc9e56dbc76b024ac97dd1f5bbee4e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-27 05:21:18.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9e9ab028203a44f505275d76e30481a7b169c7cc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-27 06:26:30.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-11-958611-1/FC/ : FAILURE in 5h 23m 20s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-11-958611-1/iSCSI/ : UNSTABLE in 2h 5m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"91093c2de47d6a0f1efdfd51b77355ad544a9329","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-27 23:09:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03c7afd4b3b3ddcc5df3934f8223332aac57d2a9","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-28 00:34:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8acb054df9e71fe1c9175dab363f426f653ae21c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-28 00:38:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"42aab75fb51589d177de5735e3806c67524f30eb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 00:38:23.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/332/","accounts_in_message":[],"_revision_number":2},{"id":"07d9ab9c93574768ba2740f3205449258595e48e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-28 00:44:54.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"be790aa43b4e98a32487e1e145c78a9781f22dfc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-28 01:17:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958611 : FAILURE in 37m 01s","accounts_in_message":[],"_revision_number":2},{"id":"cb55cfad6ab84f93e88690a7aeea00f7569cb198","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 01:35:05.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958611/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958611/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"e27418e9d3b970298cb5f2b20341ee592dddf577","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 01:40:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1727/.","accounts_in_message":[],"_revision_number":2},{"id":"39af723781425ac4c0d77caa4fd3a45fe8f5742d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 01:42:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1592/.","accounts_in_message":[],"_revision_number":2},{"id":"c7bc17e8c350396cb053acc2b251242a64f6e4ad","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-28 01:46:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ec405f61abaa4f4dafdf46d67f2d64ea : SUCCESS in 54m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bb67b0f6c7944f73beac5b79069a6faa : FAILURE in 1h 05m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d8ca12590fd27eb3c13a35c8d0ad095b2c7132c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 01:53:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42631c3a880942318f6d9f370e886136\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b0ce5a609f84b988ec3950ee8fb4f7b : SUCCESS in 22m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6838862f7c9a4565a6ed8557139543aa : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1d96132654b34dd9a4fea935e3c9ce4d : SUCCESS in 25m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"595e935c93701ac8b89174a12d00506f70a6c3e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 02:49:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64855084a4af4e19aa88c5f0c618f5d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/946c7b283d0a490b9c6bce63f3ef316c : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99ffbdbbdc934f16920791993aaab3f7 : SUCCESS in 9m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c422ea452cd48be8edd2bf496add359 : SUCCESS in 11m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/836d222915d542c88b1d816b61e720ac : SUCCESS in 14m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95d44ebef73544b39071527e46294524 : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/df26f5a9670245bd9753b876499170f3 : FAILURE in 9m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/28a4b7deb60d4a3cb387e665d2072726 : SUCCESS in 1h 00m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a27bfdd22884153966f2740a7321009 : SUCCESS in 1h 28m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3428e451ac414eb8be5aff60d1b000db : SUCCESS in 47m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/85159a1109ef4ef98ad7c5c3e0c91bed : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4390f220caad4b93a16ea8568609c4b1 : SUCCESS in 6m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1aa356cbdd64af4b58235acf2d18e6f : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ab652b736cd4854a2bc843355d2dfa4 : SUCCESS in 4m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8a746e7fff7a42fb9bcdfeaa6c543032 : SUCCESS in 8m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/db35c92f12ac4faaaf488b348eb5de83 : FAILURE in 26m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1fe0945fc8514439becdb936df13790c : SUCCESS in 4m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4496ba9784ab4871b0fe5ffc89bccbed : SUCCESS in 1h 18m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8d5934c26e23444d925a0d8f5c999a00 : FAILURE in 1h 24m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b931967fa7a14ae2be075b11830d0598 : SUCCESS in 2h 05m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4a3f9de31923499890a3e09846b410f5 : FAILURE in 10m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5e986fcf1275498e92d522de554c542a : SUCCESS in 24m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ab9cf3fa0bcc4339b08f4dbd2c086dd7 : FAILURE in 14m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/371f27fe85d7490eb062bc6ad8ed368a : SUCCESS in 1h 50m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/53881675c9b4427188accb2c80e9b220 : SUCCESS in 30m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/35a238a26d764526ba3d3eb863a4f889 : SUCCESS in 1h 27m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9421e0e82c694116a271b9a70928faed : FAILURE in 15m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7056627b20a843bea7e190a9ceef1d21 : SUCCESS in 1h 14m 13s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9519474542874287b2d49fe294e2baec : FAILURE in 16m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/892c75c7c967452ea2e9183a5dc3d692 : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":2},{"id":"48ee40f695d40af73952eb6babb599415245d4cd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-28 04:30:00.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-11-958611-2/FC/ : UNSTABLE in 2m 25s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-11-958611-2/iSCSI/ : FAILURE in 3h 50m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"add2d1def65a71b479e911390bfbd7ed6267841b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-28 20:07:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"25eb91ab0a51d3e0686c564c958d5dffcb40d567","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-28 20:09:17.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"f3238f0a8fb9dbdf090b6d1112f998bb8acf2ae4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 20:41:45.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/341/","accounts_in_message":[],"_revision_number":3},{"id":"59310b0524f6edf3eb6cdf037ddfa5927fd55d35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-28 20:55:42.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8d5e2aaad6d2362b6919f12989d488ae05b1d204","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-28 20:59:35.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"b899fc51ea1868e131d2e660661e867a77216e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-28 21:13:07.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6782149f9dd7532e4781fc089684c613ac6804ca","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-28 21:14:36.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"5111a3900d12a2c5dff726fd810eec1f4bbcf7f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 21:36:22.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958611/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958611/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"4209779118ff4ef70a18e58a19e4c93e80ec5717","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-28 22:10:47.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"bb21098583840e6a1b0f4d6640672a72992dbc91","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-08-28 22:19:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"47fd6bbb00d36de3015048f669fce348853558c0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-28 22:21:12.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"ac0ca4a2534a928957fcf950d622676382a1749d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 23:20:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c61811c3c7cf41f7814b4f76f8e6f257\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a7b06f930b3472b98e03f99a630e08d : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f79aa80e6ea4b9294b443d88088041c : SUCCESS in 11m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e2fa33522ec441c83715167698c11e7 : SUCCESS in 12m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a20e4e17d6a4e48b7b07b4996dd18d6 : FAILURE in 17m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6cbb949fd5a4b33aa7d7e95b88fd7ac : SUCCESS in 14m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/8a82aa9ce48c44d29c2090f721a09539 : SUCCESS in 28m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48d8c3da96684a3bb62bc46862d86d4f : SUCCESS in 1h 03m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04566e5ca9db42838e305352b92cdb2e : SUCCESS in 1h 39m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f2d3984751d4ba3af3e5d35a8669a36 : SUCCESS in 49m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/259e5ffc54934310ba16fb92412b24b8 : TIMED_OUT in 41m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2c3474adb2c74ae4a23ccb993c648229 : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5caa6ad255284bd5abb868e41fdf60c3 : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/de5c6b57186f4cce878bd584f3a61067 : SUCCESS in 6m 18s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6b861366fb47404a9e5ed6d2441b930b : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1c7fb1e960e84b72a553b161b4093e05 : FAILURE in 31m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f04d2b9b700e4695abf1cf89bd950dfd : SUCCESS in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/692c2d40388d438e98c2be847832f694 : SUCCESS in 1h 54m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/95fd2212a27e44928405926195711730 : FAILURE in 2h 05m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f73beeb60aa34991adc5c91c65fd22fe : SUCCESS in 1h 54m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/57f54dbb0c704dbf96df376377009cf1 : FAILURE in 12m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d41623e8746744ddacc734e536de551b : SUCCESS in 28m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ab68d4ec22a7466ca8e63f0424b2fcc1 : SUCCESS in 44m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9701e3b5dc6343cbbc9c5ee01c7a43a3 : SUCCESS in 1h 21m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e6aead3f21fc4eaf94af8bf5aee0ef2c : SUCCESS in 51m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d43764a5a1094e5a869bc159b310090e : SUCCESS in 1h 27m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f6f43b0407dc4d14af9c7495b150c03c : FAILURE in 17m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c32cdad559174ce2b1be6a9047acb1b0 : SUCCESS in 46m 56s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e47247c77985451faaddeafdd4c1adeb : SUCCESS in 1h 23m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54c45d7e74154c2793f1c45941b6ef05 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":5},{"id":"65c5c8c4bb3dc484875633e6abf8a8fc455d45a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 00:20:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f290ec9fb944fdadf114bc94f3b81b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/60a8db91ef56466b8b4ee0961be1ba3a : SUCCESS in 27m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4546357a105347e180e3821cabd39bf5 : SUCCESS in 25m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a8c3564d82844aedb8b996dade5bb7c2 : SUCCESS in 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4ae4f0ff91eeea0660ef3e29a7145d4a85a061fb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-29 00:21:15.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/345/","accounts_in_message":[],"_revision_number":4},{"id":"6ead6d19103ffec3019a37b5e06a5497df25a80d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-29 01:14:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1605/.","accounts_in_message":[],"_revision_number":5},{"id":"f996a300675e431dbc47d4e95552459681749cfa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-29 01:16:59.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958611/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958611/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"8973004aac98e38af678e2c4f6c91f816ecfb63d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-29 01:17:00.000000000","message":"Patch Set 5:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/346/","accounts_in_message":[],"_revision_number":5},{"id":"37cbd73e3ab7dff6ed1bd18856a2c0be4a7694b6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-29 01:18:50.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-11-958611-3/FC/ : FAILURE in 5h 8m 0s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-11-958611-3/iSCSI/ : FAILURE in 3h 41m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"78176311137fd21e74d156049312b37cff96d36d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-29 01:23:20.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"58eef4637fae123d1c0cf90b4bad2bc611cd71b1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-29 01:32:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1739/.","accounts_in_message":[],"_revision_number":5},{"id":"d593ac08576fd2e1b9d26564a80c2bd2fce0adab","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-29 02:08:34.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958611 : FAILURE in 36m 42s","accounts_in_message":[],"_revision_number":5},{"id":"9eed1d7a191e9a0da978339357baa60101a0bcb8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-29 02:13:11.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958611/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958611/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"b8c7de3683851f116a2c44b2c3f44f359cd1996b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-29 02:17:11.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"cbe61e2a80c71dc66e6ff05df63633b1cf6d3bd9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-29 06:41:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8f7a425a844d4f23a69624a121ae87b3 : SUCCESS in 1h 38m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/14fa3c42cc1d44c1a0276b27921b54b7 : FAILURE in 1h 49m 45s","accounts_in_message":[],"_revision_number":5},{"id":"6d06f6aee108af02a97100068c59361f6140b45b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-09-01 14:42:14.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/958611/5/20250901192032/NEC-V-ISCSI : FAILURE in 3h 28m 50s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/11/958611/5/20250901192032/NEC-V-FC : FAILURE in 4h 21m 40s","accounts_in_message":[],"_revision_number":5},{"id":"7309e50161a93be98a3df62d19ba8809707bcfcf","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-09-04 20:20:50.000000000","message":"Patch Set 5:\n\nCI results  \u003chttps://github.com/gamboa-hitachi/Hitachi-CI-Results/tree/main/958611\u003e","accounts_in_message":[],"_revision_number":5},{"id":"5b5c24bfbbbf3b97455188d92303237c8b1717c3","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-09-04 20:41:29.000000000","message":"Patch Set 5:\n\nCI results  \u003chttps://github.com/gamboa-hitachi/Hitachi-CI-Results/tree/main/958611\u003e","accounts_in_message":[],"_revision_number":5},{"id":"2225d04cb29ba9054ee9880759f206b6528401c3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-09-22 23:43:36.000000000","message":"Abandoned\n\nincorrect relation chain. re-opened as 959901","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7fbaa61a0de15a8fb14936b824109cd003086e6f","revisions":{"b259f9e6396ed820b145bb7173dcd8cb3f3d9598":{"kind":"REWORK","_number":1,"created":"2025-08-27 01:02:18.000000000","uploader":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"ref":"refs/changes/11/958611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/958611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/958611/1"}}},"commit":{"parents":[{"commit":"c27af2538431057d2c3f5dca919a850ef38a0f8d","subject":"Hitachi: support extending volume has snapshot(s)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c27af2538431057d2c3f5dca919a850ef38a0f8d"}]}],"author":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-27 00:59:29.000000000","tz":-420},"committer":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-27 01:01:12.000000000","tz":-420},"subject":"Hitachi: add support for Hitachi VSP One B20","message":"Hitachi: add support for Hitachi VSP One B20\n\nChange-Id: I748713c1ac1faece0865ebcd9a273a94d6d0b625\nSigned-off-by: Anthony Gamboa \u003canthony.gamboa@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b259f9e6396ed820b145bb7173dcd8cb3f3d9598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b259f9e6396ed820b145bb7173dcd8cb3f3d9598"}]},"branch":"refs/heads/master"},"689c42ba9e4189404f3d7e768bf2d2ec14ba7c68":{"kind":"REWORK","_number":2,"created":"2025-08-28 00:38:15.000000000","uploader":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"ref":"refs/changes/11/958611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/958611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/958611/2"}}},"commit":{"parents":[{"commit":"c27af2538431057d2c3f5dca919a850ef38a0f8d","subject":"Hitachi: support extending volume has snapshot(s)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c27af2538431057d2c3f5dca919a850ef38a0f8d"}]}],"author":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-27 00:59:29.000000000","tz":-420},"committer":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-28 00:38:07.000000000","tz":-420},"subject":"Hitachi: add support for Hitachi VSP One B20","message":"Hitachi: add support for Hitachi VSP One B20\n\nChange-Id: I748713c1ac1faece0865ebcd9a273a94d6d0b625\nSigned-off-by: Anthony Gamboa \u003canthony.gamboa@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/689c42ba9e4189404f3d7e768bf2d2ec14ba7c68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/689c42ba9e4189404f3d7e768bf2d2ec14ba7c68"}]},"branch":"refs/heads/master"},"163243060d9929b19b8b82505e3e85cb7e02957e":{"kind":"REWORK","_number":3,"created":"2025-08-28 20:07:44.000000000","uploader":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"ref":"refs/changes/11/958611/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/958611/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/958611/3"}}},"commit":{"parents":[{"commit":"c27af2538431057d2c3f5dca919a850ef38a0f8d","subject":"Hitachi: support extending volume has snapshot(s)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c27af2538431057d2c3f5dca919a850ef38a0f8d"}]}],"author":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-27 00:59:29.000000000","tz":-420},"committer":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-28 20:07:27.000000000","tz":-420},"subject":"Hitachi: add support for Hitachi VSP One B20","message":"Hitachi: add support for Hitachi VSP One B20\n\nChange-Id: I748713c1ac1faece0865ebcd9a273a94d6d0b625\nSigned-off-by: Anthony Gamboa \u003canthony.gamboa@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/163243060d9929b19b8b82505e3e85cb7e02957e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/163243060d9929b19b8b82505e3e85cb7e02957e"}]},"branch":"refs/heads/master"},"c27df2a8904eb9085cf208722bf58132e3e64469":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-08-28 20:55:42.000000000","uploader":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"ref":"refs/changes/11/958611/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/958611/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/958611/4"}}},"commit":{"parents":[{"commit":"72f1bdcc014f38b3af2eba8a8eef5ed28b5b4685","subject":"Hitachi: support extending volume has snapshot(s)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72f1bdcc014f38b3af2eba8a8eef5ed28b5b4685"}]}],"author":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-27 00:59:29.000000000","tz":-420},"committer":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-28 20:55:31.000000000","tz":-420},"subject":"Hitachi: add support for Hitachi VSP One B20","message":"Hitachi: add support for Hitachi VSP One B20\n\nChange-Id: I748713c1ac1faece0865ebcd9a273a94d6d0b625\nSigned-off-by: Anthony Gamboa \u003canthony.gamboa@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c27df2a8904eb9085cf208722bf58132e3e64469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c27df2a8904eb9085cf208722bf58132e3e64469"}]},"branch":"refs/heads/master"},"7fbaa61a0de15a8fb14936b824109cd003086e6f":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-08-28 21:13:07.000000000","uploader":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"ref":"refs/changes/11/958611/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/958611/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/958611/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/958611/5"}}},"commit":{"parents":[{"commit":"72f1bdcc014f38b3af2eba8a8eef5ed28b5b4685","subject":"Hitachi: support extending volume has snapshot(s)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72f1bdcc014f38b3af2eba8a8eef5ed28b5b4685"}]}],"author":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-27 00:59:29.000000000","tz":-420},"committer":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-08-28 21:11:19.000000000","tz":-420},"subject":"Hitachi: add support for Hitachi VSP One B20","message":"Hitachi: add support for Hitachi VSP One B20\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/869991\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/850830\nChange-Id: I748713c1ac1faece0865ebcd9a273a94d6d0b625\nSigned-off-by: Anthony Gamboa \u003canthony.gamboa@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7fbaa61a0de15a8fb14936b824109cd003086e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7fbaa61a0de15a8fb14936b824109cd003086e6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
