)]}'
{"id":"openstack%2Fcinder~958707","triplet_id":"openstack%2Fcinder~master~Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c","project":"openstack/cinder","branch":"master","topic":"bug/2121557","attention_set":{"37888":{"account":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"last_update":"2026-01-23 12:34:53.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-04-02 19:43:48.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-08-28 23:38:09.000000000","reason":"\u003cGERRIT_ACCOUNT_35467\u003e replied on the change","reason_account":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-01-23 12:34:53.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-01-23 12:34:53.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"35467":{"account":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"last_update":"2026-04-02 21:03:12.000000000","reason":"Someone else replied on the change"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-02 19:43:48.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-04-02 19:43:48.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-08-29 15:15:28.000000000","reason":"\u003cGERRIT_ACCOUNT_35467\u003e replied on the change","reason_account":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-08-28 23:38:09.000000000","reason":"\u003cGERRIT_ACCOUNT_35467\u003e replied on the change","reason_account":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"}}},"removed_from_attention_set":{"13915":{"account":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"last_update":"2026-08-05 13:43:58.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13915\u003e by clicking the attention icon","reason_account":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}}},"hashtags":[],"change_id":"Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c","subject":"Improve restoration of encrypted volume backups","status":"NEW","created":"2025-08-27 22:35:49.000000000","updated":"2026-08-05 13:43:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5b264ab2871c9b3c3e2ebdc0c767a1d6acbe7e89","_number":958707,"virtual_id_number":958707,"owner":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-02 22:04:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-27 23:08:06.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-28 00:03:09.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-28 00:05:18.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-28 00:15:50.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-28 00:22:11.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-28 00:38:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-28 01:20:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-28 04:42:49.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-08-28 19:43:59.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-08-28 23:38:09.000000000","updated_by":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-08-28 23:38:09.000000000","updated_by":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-08-29 15:15:28.000000000","updated_by":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-11-21 15:56:03.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2025-11-25 13:32:59.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-12-19 09:50:56.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2026-01-12 20:18:12.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-01-23 10:28:06.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"state":"REVIEWER"},{"updated":"2026-01-23 12:35:00.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-01-23 14:15:13.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-23 19:59:52.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-01-23 21:35:31.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-24 21:46:20.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-02-04 15:39:14.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-02-20 14:43:08.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REVIEWER"},{"updated":"2026-02-20 16:58:29.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-02 21:03:12.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-04-02 21:50:16.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-04-02 23:09:05.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-19 19:50:05.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"c34ce0cdf8413f7e1557a37d2a426641ac3c7c34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2025-08-27 22:35:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19871380f4a516e80588154d36a8da9c914d9d1f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-27 23:08:06.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/331/","accounts_in_message":[],"_revision_number":1},{"id":"9e5cd99104b8c6507080fecd51666e408832ee1a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-28 00:03:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958707 : FAILURE in 1h 25m 52s","accounts_in_message":[],"_revision_number":1},{"id":"895038bf52ce6661783e6ffe8f1ed114bafe80aa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 00:03:49.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958707/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958707/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"4fc612a7b701816a7a6efadef6d7ddff2cfe18e8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-28 00:05:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2e8f5fcbc460423593c0129e4eed2d6f : SUCCESS in 44m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d8fc6c2863144369a73efc834f2af2a4 : FAILURE in 48m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c719f36f15e8f8723e5323949b9cdb55c04834ad","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 00:15:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1591/.","accounts_in_message":[],"_revision_number":1},{"id":"930080d20c69aacde0a8d2de213b67d9bc719716","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-28 00:22:11.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"aaa8036be5097b11e55619e24cfe269fde41f15d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 00:30:05.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1726/.","accounts_in_message":[],"_revision_number":1},{"id":"2db90f6d52dd0562a816886b9f2247bc78df9572","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 00:38:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd6e346a3c9746ffb13411d80d4e783a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/99a1ee77a7944b779ea498b11f66de74 : FAILURE in 24m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/65618e6d6cca48249d5b2ced60070e4a : SUCCESS in 25m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dab3b369235b4f519c4bea5f88db6ebd : SUCCESS in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f37dbb2a35fda1159b68d66ab69d3d27a096dfe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 01:20:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e125ccad6de4e35b55d8a1f48ec1f4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdcb797667e34a1ebcdec0063112a1d2 : SUCCESS in 6m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/caef00b2225145d9bb34b38dd08a854a : SUCCESS in 9m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7641423b78704db1835ba095f755115c : SUCCESS in 7m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8faf9b1f33c9448694af1ba974089b04 : SUCCESS in 14m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd78b4b027cd45b58d2dfb0259d5c906 : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/a262d5e130c246ea9bf2d4d4c7597b7e : SUCCESS in 58m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4aef8f8f543e48b990eb809dafdd71bc : SUCCESS in 45m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ac4221ebc311468980106f56ecd44d97 : SUCCESS in 1h 25m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a617bf4d13f747a68fafe740487461fb : SUCCESS in 59m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3f29987a43d14395a7799c5e1b986512 : SUCCESS in 20m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5accec177bf6404a9dcf5811665aff2a : SUCCESS in 6m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c8ad885d20b44d388643257d2d300b1c : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3120fb2f18f842a695cebe9107585599 : SUCCESS in 6m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/baf6d39bc81942fc99cbe11404db8b40 : SUCCESS in 9m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/767de971c9934757b3b3b4e46fa39558 : FAILURE in 37m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9d44ce85d3d74974b70ca37f08569c4a : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f6cf29d74b2342f4bcb8d76333a0aa52 : SUCCESS in 1h 45m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b1f88092aa7a4b7a82b0120b679fdfd4 : FAILURE in 2h 37m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/023c26ba9e59458f93b110e9c9eed613 : SUCCESS in 1h 51m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a63ef1c997304cc2ad6b0dbf587b982a : FAILURE in 15m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/07c26dfbf97c4dd6a3b18632eb48fc2b : SUCCESS in 17m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/391ff05719fe478cbecf619b08f78689 : SUCCESS in 57m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/432fe346ef1f4bcbadbc170f7c710cdd : SUCCESS in 1h 08m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4a292761f4cf40cb971bd66cd75fe2f5 : SUCCESS in 40m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eea4d400ac3640ba996bef2d1345695b : SUCCESS in 51m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/632c0f202db54885a8cd43038a5c6dcd : FAILURE in 15m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be8c9c8e325b461fb82fe9ef2c9a6a71 : FAILURE in 51m 13s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/908e70259281489493c9d5f12ec8ead7 : SUCCESS in 1h 56m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ddfc5bbd35e4bba91bc9f338e47d671 : SUCCESS in 33m 57s","accounts_in_message":[],"_revision_number":1},{"id":"dc46d121d7c5faf1819714ea065026ffa2173448","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-28 04:42:49.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-07-958707-1/FC/ : FAILURE in 6h 5m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-07-958707-1/iSCSI/ : FAILURE in 3h 46m 0s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"2084c390388bbac1e565267e45f0f9769c02ec7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2025-08-28 12:34:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c7456256e29dcee7db6585b234f2caed2b2ae2d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 12:34:40.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/338/","accounts_in_message":[],"_revision_number":2},{"id":"90aa684945089f22237e41fc4150baa85e59238c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 13:27:29.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958707/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958707/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"817b01ec1e187867a1273be546d6f17ed08769d2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 13:34:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1598/.","accounts_in_message":[],"_revision_number":2},{"id":"8a8047705702fca532efa01fb339f10b05bdc956","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 13:35:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1733/.","accounts_in_message":[],"_revision_number":2},{"id":"9011b0defc92cb0e8af8c74c548682496586bb21","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-28 14:01:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958707 : FAILURE in 1h 25m 07s","accounts_in_message":[],"_revision_number":2},{"id":"287a09dccf8ab2562bdaef71f30bf49ccc09e768","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-28 14:16:38.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"bf7d23fb9d056b10d02d5778fa44d3dfc0e6f526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 15:18:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc693d6db26d4073bd01c6e7d0d4b940\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cc08f13627a4e4c95aaf413e1cf6144 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/984c101f5ce9461b94aaab278ad653bb : SUCCESS in 12m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c4b0b09d7b4425da0f84c3840e9b268 : SUCCESS in 8m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f687209b4354e1b91cbb4c735177f2e : SUCCESS in 12m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0932453c28242e38bc56bc4e56597ae : SUCCESS in 14m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/cf98e52b4dac4bb29578a728b471d507 : SUCCESS in 53m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/26a7763ec67e4ba6a90f6df0d201eca2 : FAILURE in 40m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6cc1e76d6455466ab10b026916bbfa48 : SUCCESS in 1h 40m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdd9a551aa1643f7a1438b0d36299034 : SUCCESS in 1h 05m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a3cc8cea30bb46278b8b3cac8042323a : SUCCESS in 21m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/30e38add4d5d4b93b204f98d4d409dbf : SUCCESS in 3m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/47009d83d5954961b16e6dea1cd73d6b : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/65428b8589214a4c8d80ff01e87be7e0 : SUCCESS in 5m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b8633b84b04d411091bcf27ec96a7812 : SUCCESS in 9m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5c292a56b4b94e439dc5fc9920eae7b7 : FAILURE in 22m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c8bd4c0771404784a0e56a907b0cb7d2 : SUCCESS in 6m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3c0e9b190baa446ba1726215dae8a336 : SUCCESS in 1h 46m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1a99d64a7c4e4a7a9b83666e014ff7e3 : FAILURE in 2h 36m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ec3b4af756d84981a6a31b1cb8cc9831 : SUCCESS in 1h 30m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9dd9a59b554246eb8f9acb4fb0cc1b10 : FAILURE in 17m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/33af14e4132c4429aee1451bba67a8ea : SUCCESS in 34m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cd8f9003be7344d48cb3723a8bc8ecb6 : SUCCESS in 51m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/700ef83726024cb59c6bc433bd509deb : SUCCESS in 1h 34m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1ff7bcdcaf5042a8812206bcfe60989f : SUCCESS in 37m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7a40ef74c2b240629287e0168047fe55 : SUCCESS in 1h 22m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d7540628e2a84a8289778abba8bdd69e : FAILURE in 9m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/61d63482a78d4ced89f296bd8a0c968c : SUCCESS in 1h 32m 30s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcf2ff20d33a49eca8f7097cbac93acb : FAILURE in 9m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09a54ec269374280b288f82afa5fea81 : SUCCESS in 31m 22s","accounts_in_message":[],"_revision_number":2},{"id":"3ff8773965cf916d791056537304c1eb8f5c2fe1","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2025-08-28 15:40:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9fafb8b21b9d309c463dd9cd7b4abeb46e6a8a13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 15:51:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b58b24260ef42e583e65d9cdba1b0f0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c27ccd14fd524cc99bcbc32ae2910cdd : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/048b3a972aa244cbb936d6ff72733f4c : SUCCESS in 31m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f4ca63e6e44a4dfca37a01b2fdc7f923 : SUCCESS in 26m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4836a854748b3d3244b8ecfa73696c65722ddc63","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2025-08-28 17:40:57.000000000","message":"Topic set to bug/2121557","accounts_in_message":[],"_revision_number":2},{"id":"77bba6faa5a92840d23ba7422c92b1ea1c1b90c9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-28 18:05:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fd088cfd56ee4579a7d106a9e0a72bb6 : SUCCESS in 1h 42m 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/163af93cf00b475ebfa52c08c6e912e5 : FAILURE in 1h 46m 03s","accounts_in_message":[],"_revision_number":2},{"id":"918b9d36653a2ba1b25623a914c4c63546a066c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 18:10:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a3848b1cf26415cb2dab0b6adbf4256\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e67584602598482d9b238a7be4c9bf74 : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72fadef944294965bbf84433a357cf43 : SUCCESS in 9m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a3ebf3bb5b140b494916ee4cae7927a : SUCCESS in 9m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f9b8a69eac642308fab77b55dc97ebb : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa7017b726c54fc2b1e8b1c73307d11a : SUCCESS in 13m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/a989a2e797d2476a9758a66fc7adc091 : SUCCESS in 59m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2fce6e23862a438688d737718a474b2a : SUCCESS in 47m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8857dbce4634b41adcbcdbcee46a541 : SUCCESS in 1h 47m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb7c7f27a91f4e51bb6986688d7be359 : SUCCESS in 57m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/90dd4bd360dc4664b0e1f43ea90c93b6 : SUCCESS in 24m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f3c53b260bc94d8193703a536b886d36 : SUCCESS in 6m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/80fc99135e6049d99d93feef03b47d39 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/618db4ab830a48629627a41a169f2ed8 : SUCCESS in 6m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0e71cb4ee8a2403f8be5ba735b63adac : SUCCESS in 8m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c5cd94078f6c49d38e0d308545ab9879 : FAILURE in 29m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ceea5fcf2bff441ab1d39b8f7018a82c : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2bac12685716402b8473f3804f9a7381 : FAILURE in 1h 08m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3b01c58781174141b4ef68aead21c252 : FAILURE in 2h 24m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/886177c90113404f8b13afa452b54ca9 : SUCCESS in 1h 36m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f9bd3d75f48e423a9ec46d167a45780a : FAILURE in 15m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2fe7b1bac5984931b89604f52218c683 : SUCCESS in 38m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/34f859881d5241f18a3ff4cc8169e78a : SUCCESS in 59m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3ec83c564de74ecb9d33581d0b21da55 : SUCCESS in 1h 38m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/96f8ee53a2074b9692ce079757c314c5 : SUCCESS in 41m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/67d7288a36c4407489eec43f65ebd9ba : SUCCESS in 1h 33m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/743c206cb2c94621b5073b1831e9345a : FAILURE in 11m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c7fecf24e86d4081a2d9445c07ce6e7b : SUCCESS in 1h 44m 06s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/721ac9bb575741cf9260640192fcae5b : SUCCESS in 1h 27m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19734cbc00d04b2686bd2d2239eedbc8 : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d841f7d9a7a76ac5fca463555ee8d15f91e0ac03","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2025-08-28 19:42:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd637e674ad1a128b73d4483d2261c0150be8b20","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-28 19:43:59.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"b0c468955e842f5504b18a2c9aa40a915c2d4f3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2025-08-28 20:19:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ccaa267cc0669a7f9bd61d8b20ed7f32c27c955a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-28 20:20:52.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"ad956f405381435e0238fabfaf25fcd30ac2f9f5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 21:36:24.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/342/","accounts_in_message":[],"_revision_number":3},{"id":"1f36c0a650e6165ab8761d500d448180c6032c40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 21:54:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6232fef177b24a70ba06c179ae95b3b6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1df91f5df0f24785bc2902d69c5471a4 : SUCCESS in 27m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d9c5575fa94412ea8f146b75bd1aa21 : SUCCESS in 20m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5bfe8c984c784009aa9585e4f8cbc0d3 : SUCCESS in 30m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5498d791963e66535a03a5ef71c3143e4c6c68f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 21:59:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1736/.","accounts_in_message":[],"_revision_number":3},{"id":"19b3206e040aac0305040a766bce3456308293b3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-28 21:59:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1602/.","accounts_in_message":[],"_revision_number":3},{"id":"fcafa260d30e879c67361480092285e2e8eae69b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-28 22:30:57.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958707/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958707/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"4c727f5cbb74ff2155835a783fe0f636200548b9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-28 22:35:43.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"0ffb9aa7e074b4f62cb97b4177d5e09e92143a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 23:02:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff916a2e29c04a2fa59f473611db02c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad0219b3ed3f43ad8d72720364ddd240 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b44be9360d44fa99099ea0daa7fc252 : SUCCESS in 11m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dfee90169f194dbc88de046af50d5b48 : SUCCESS in 12m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca423f64a7ce4a27881c30b35cbbb5f5 : SUCCESS in 14m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74fc6ecbd0a34636bbfb3894b846ad75 : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/cc95d76957364c49853a645725b4bd96 : SUCCESS in 1h 02m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d71354fb1b1b422bbac8c9c1e07c5b84 : SUCCESS in 1h 00m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/176b8c676e8f42549095e7cd4a8e7caa : SUCCESS in 1h 27m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11b7b4594f08499c8e60259a6ba208df : SUCCESS in 1h 01m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1f035fe617874660bbe11f5dba4c239f : TIMED_OUT in 41m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a5d47d67fe6947e0a335b42111a5f24d : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dc840f0b67ec43eeb23f830108a4c5f6 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/38231c89e75a45d2b34e2ea7aaaaa822 : SUCCESS in 5m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2e7a37a5f37b4f48a748fcf9c43bdb42 : SUCCESS in 8m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ad5c1bebe082457a976128a0c4664157 : FAILURE in 36m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/85358386d6d34b77bc71e2caa788c824 : SUCCESS in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4745cf5731dd4617b39806283a02134e : SUCCESS in 2h 01m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4d0e232e31ac4a4caf716fe1cff17552 : FAILURE in 2h 38m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2e0e1ca10e6b4024a35d2536d5934744 : SUCCESS in 1h 40m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5a1574bf1bfa4cf39b14cf1db4ec08bf : FAILURE in 19m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/20d73bbc9a2b463781b12a60d4002213 : SUCCESS in 35m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b9084b3e6fea41648f4732864d5d35fb : SUCCESS in 34m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/103022a9f55c4765bfa1ee3a9d3c718e : SUCCESS in 1h 22m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/da1cf635c3714ffa85c62552d3820d18 : SUCCESS in 30m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4c8d5023907c4819a544658b598b45ee : SUCCESS in 50m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/72924d5784e3437087b4a78a629c7d28 : FAILURE in 8m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/88f481c0e3194069a908d1a15512736c : SUCCESS in 1h 47m 30s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ab14f25e26644dd924575f09430ce0a : SUCCESS in 1h 29m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6992f7d91a8e4632a25474d462919c87 : SUCCESS in 44m 15s","accounts_in_message":[],"_revision_number":3},{"id":"326b1b2d7c3d8ef976319068e1ecec6997c57fe3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-28 23:19:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958707 : FAILURE in 1h 24m 33s","accounts_in_message":[],"_revision_number":3},{"id":"3d4d2ca9fed19bb3ad4cb0495addd9a20fc43437","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-29 01:21:49.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-07-958707-3/FC/ : FAILURE in 5h 39s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-07-958707-3/iSCSI/ : FAILURE in 3h 49m 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"2c09a8f5615f2d0b3234cdd290b0810305fb0b4a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-29 01:23:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5ce0810f7621475da64f8fdf4b699ab5 : SUCCESS in 1h 35m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c5868c05c0ff4735a3779fcc2b16e5bc : FAILURE in 1h 50m 00s","accounts_in_message":[],"_revision_number":3},{"id":"e95612dc52523e2d942c37416bc30dc6f1c38a5e","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-21 15:56:03.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e0d68458be3cc6091c682b7c93516f3c575c06e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2025-11-25 13:32:50.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"83b138b1192b092b8a71d2da563bb66ed294cc58","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-25 13:32:59.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/67/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"e9085f80e7b5838fd7d6ce3047b1c793d00ea322","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2025-11-25 13:39:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4f7de8c98b180e00db43ddd1e212999c201c6f68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:11:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc8cdbb7aa674071bec157f392f9a662\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f08cb070d51541c39aceba4dde84094c : SUCCESS in 31m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1292bf1b978c460fa6982f3c1bd7c1f9 : SUCCESS in 30m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/42d9904120e244c1a7b85c1ffadd3156 : SUCCESS in 35m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6f821bd398fb956fbc160fecbfea07254c1397e3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-25 14:29:19.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c .","accounts_in_message":[],"_revision_number":4},{"id":"14b8bb35b75ef04376e4c59722a12c2d8d5a9496","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-25 14:43:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3765ff0c6dea403ea19eb64170da099b : FAILURE in 1h 06m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dfab9456d629492281b4e1cdb4dcd3e0 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":4},{"id":"b323d0293912576db5d622d7d9efa996d6ff0919","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-25 14:44:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958707 : FAILURE in 1h 10m 27s","accounts_in_message":[],"_revision_number":4},{"id":"0c748ff84a1d623e8279dde8ab58b850d19d484e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-25 14:51:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2721/.","accounts_in_message":[],"_revision_number":4},{"id":"84f6d54063c76627ee7276ca25c6fe042cdaa444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 15:54:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c957883a49645388d5ecc07d40d6796\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c68c85dd66a4949b0d30cb8d1d7278b : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2932029820634072bca39e112942f9a3 : SUCCESS in 10m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f5aaa9ccdb4421fa6df165ab574cb0d : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12c556e2cf8467cafbe46851ea32b5d : SUCCESS in 15m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/a49011bfa6e74cb48a25ed634b9a5e84 : SUCCESS in 1h 02m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad66d37e2244431fa2c40d29c97f438e : SUCCESS in 1h 02m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f84c481b9e94e7a86405545eda508ae : SUCCESS in 1h 40m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afeaef49cf45479382c7957afee86b6d : SUCCESS in 36m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6021f3a4dd654bd58ce332d441561a82 : SUCCESS in 16m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a89df7b76079426f89d69bef17eeada3 : SUCCESS in 21m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da8a5b4d0c12411b9a77628f9f48a4a9 : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff05fcb90bfe49b68b79491ebfd3ad17 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5278a59fd87144dcbb5713233f59e6a8 : SUCCESS in 7m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f6e214706b38471fa86abb7e72271443 : SUCCESS in 8m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/196115825c0546539cb8e50f3d62a3d9 : FAILURE in 38m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c13dcb9ea5fe4fc4bd21d9bb23645914 : SUCCESS in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9a332ba02e944c91bc0476b318d508c9 : SUCCESS in 1h 05m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a976dfbd35a0456fbc2b75b96854051e : FAILURE in 1h 54m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/baf7cc3bcca545d1b2b895c48c2e7f5d : SUCCESS in 1h 46m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ce14f52707484dda812d91c0bcb2c8aa : SUCCESS in 2h 17m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ec30e0861ef641c5bba0e22fa0b0d9c0 : SUCCESS in 34m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0857025d11434af8a490dffc57b7c5b0 : SUCCESS in 1h 12m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/856a29a54ef84d1a80aae08e404ae7ed : SUCCESS in 1h 41m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e321779d6f414c608a1286168d843dea : SUCCESS in 44m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d8239c4a4a8141ddb32d11735f0ed19b : SUCCESS in 1h 52m 41s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d7a97bb887f541b69ee0a2bb89b27bcc : SUCCESS in 1h 30m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d28e4669dd324ac684c8f7d66ce2f775 : SUCCESS in 1h 28m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20628ca55a204def91001368c657ddcd : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":4},{"id":"00ec562964cffd5f8a851de4a8b5f492e736c3f2","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-12-02 06:49:00.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f52c718bc619b632027ae2b5750da407b9979357","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-12-19 09:50:56.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c4a0b4fd8363e2462b83cfbec1907d3283bea221","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-01-12 20:18:12.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"65d441c91570a610b84ab22c83c99ea4155fb624","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2026-01-23 10:28:06.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0d311fb50b93735368b67f1a4921ea3fe16b690b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2026-01-23 12:34:53.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6ef88b8eb056f5edc9438dd9e1b1a44509f051d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-23 12:35:00.000000000","message":"Patch Set 5:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/660/","accounts_in_message":[],"_revision_number":5},{"id":"20d61597821af6641142acf303b854d9c325dd9e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-23 12:35:04.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958707/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958707/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"482b131592a319cd59e7686a9d8b0951fd882dd0","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2026-01-23 12:38:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9af0caf9395400c0a26c381f0a2717c43bc0ef96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 13:08:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d92264e7175468dbbc1866515855fc9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be82c74b55ba40fa9256d0f472900b6a : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8200638c06c745c19473f7e71d99791f : SUCCESS in 22m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9ae716cdf1ae467d881497d9676bef73 : RETRY_LIMIT in 6m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c71b125bbd8e0741d20e2a49b76c3d02d4609cff","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-23 13:31:06.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c .","accounts_in_message":[],"_revision_number":5},{"id":"46f66f483d14285da2696b08bd673ea4ac81019d","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-23 14:15:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/f644a3b10f5446f588ed505ce3407ffa : FAILURE in 45m 38s","accounts_in_message":[],"_revision_number":5},{"id":"06bb8aeee1de9c0abb37481180bd81864447b101","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-23 14:55:08.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958707 : FAILURE in 1h 23m 08s","accounts_in_message":[],"_revision_number":5},{"id":"74065708f1909538f63fd81a6afe99b255d8be54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 15:15:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/536eefda73a6496ea24fd41036378915\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2ac2339d23249cbaa48e2b6e5138e52 : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b65a5608549d4f2f91c9e1534f2d789d : SUCCESS in 12m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c93f81aa6bf4ed4bd01dd2880bdd171 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b987aef963049de9b042c438ee462a9 : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/6cac8813065d441abdbf2de911fdc972 : SUCCESS in 1h 00m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/943be0fe6a8d44f2a14eef1e2ef19194 : SUCCESS in 1h 04m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd8bfa57cb8c4f33877392d76e5c4e59 : SUCCESS in 1h 27m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae1079d3dedc4acd88edc2e4e645d942 : SUCCESS in 59m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6ff368b20d948b1ae8055f29e66a2bc : SUCCESS in 17m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8ec7af611ba24e82bad166532cca5505 : FAILURE in 26m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/864b5f84a210480b97abd23329c0a751 : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/df117ab9727c47538155ab81b8342169 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5bab8bba4034043827b85e865502e40 : SUCCESS in 5m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cf489420015f4307b8501e3eb7e54025 : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9d295ccf425b4d08a550e443a4a99c90 : FAILURE in 36m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/71f254c444644dafa108d7f56bb32934 : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/13a8c42abc4641b6a808cf7e48e17da3 : SUCCESS in 1h 52m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c6eedc97e85444d984e306521508702e : SUCCESS in 2h 34m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2ec34a0bafc045cda104a57dd6da5d0e : SUCCESS in 1h 58m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bb4d4f7ca78e425082a3f02fed15bf77 : SUCCESS in 1h 49m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2cc23dc6b9cd420ca3647cace4cce55b : SUCCESS in 35m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/554e257549c14c3ea32714fcd59aeaf5 : SUCCESS in 1h 08m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1d0b83379821408db87e0c17b859447f : SUCCESS in 1h 29m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4b938dfb654341c5852ad81fd2c9af27 : SUCCESS in 33m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1cccdf1b6ec5422bb8617909c2416d94 : SUCCESS in 1h 51m 28s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9829bcc8946a4443aca72d99ab99fe1b : SUCCESS in 2h 09m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75550baf2e5746b88aa0951083c554a9 : FAILURE in 1h 33m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d69682edfba43c788844a617b64cab3 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":5},{"id":"08573922709abfe05cee763db571934bb0d447df","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-23 15:47:38.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f8393873d0ae488a94812e11c1ace99b : FAILURE in 49m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9c070286078945d5abd04352aca08223 : RETRY_LIMIT in 15m 44s","accounts_in_message":[],"_revision_number":5},{"id":"465bd4a0db32ca6d84dda356abd66aadf620054b","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2026-01-23 16:36:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bb98ad563c4a2ea2c5fbe818fd3379eb3e8f6d24","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-23 17:39:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ac7b2677ce584f34a2c220d0173b803e : FAILURE in 45m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/443837c8ec0e4d1589dd8588921a140c : RETRY_LIMIT in 14m 49s","accounts_in_message":[],"_revision_number":5},{"id":"956380749a2bd4d2293208fe552fcbb8e0f1c6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 19:52:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90c7a94b45244f8bb73f74d66b1db825\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01e49721b7a94cca806fc6b9c3e2ef2f : SUCCESS in 8m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a60671476af9474dbd918acfce6e3974 : SUCCESS in 13m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe18e5b79fb54266a2ac3e668a7da833 : SUCCESS in 10m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/359d0314f1234a8882e8df6486cba55e : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/3abe3b4d0faa4f2d9a0d35ac13ff5041 : SUCCESS in 1h 00m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3cfac1114ff3456882d14297f6b0bd8a : SUCCESS in 1h 11m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c6e356c75ae4adaa03da11223970cc5 : SUCCESS in 1h 16m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3eac8a80b50486b87af4b9132e31b97 : SUCCESS in 56m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ee6d90d616e441fbfb98fe75465314e : SUCCESS in 16m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca1502d399234e6685dee0235f22b5fb : SUCCESS in 20m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4e7226f9e3cb401c9b3efcfffa735bee : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/edb0884b0bf948c981ec3208d81fd044 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/63920c93d2894923b6a8fddbde30f4b5 : SUCCESS in 6m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/50ab2dddd16e473a96c907ee9f4375cd : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8b4611583c384be5b218c5f72ceb7f8a : FAILURE in 21m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dd0e5781cd7f4fdab43643d538fff325 : SUCCESS in 3m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/81e63931a8754bafa6b3553c0b44565f : SUCCESS in 1h 56m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/93b85f22cd9b4934902694c75842bf35 : FAILURE in 2h 54m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/10c63faa661d409c8a7c9bf67c568355 : SUCCESS in 1h 59m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fbd3de8aba4f4026b189293996016ad3 : SUCCESS in 1h 58m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5849bde4bf4a44db87b68567c5e6883a : SUCCESS in 21m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c3f0286d702c40a28c2464b29ae34966 : SUCCESS in 43m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2a21e76f38be49569096903b8c1f7f58 : SUCCESS in 1h 43m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c286e3a882c742a2bfd13fea40e6bc7f : SUCCESS in 41m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8c020e2b28594bedb0a2819fc0871951 : SUCCESS in 1h 44m 25s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/a1403596864147ecae540bb4b399492b : SUCCESS in 1h 50m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5ab5a960f0c74009a33090ef0c9a1118 : SUCCESS in 57m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6749d520dbc435180be3cf059c9bbae : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":5},{"id":"0d4bff6d91d7cdff1463bbdb3ddee5f5d401351b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-23 19:59:52.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_01_2026_19_53_09\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"2ccadc7d971bbde376c7e7254f2aa072f328224a","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-23 21:35:31.000000000","message":"Patch Set 5:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_347/index.html : SUCCESS in 98m","accounts_in_message":[],"_revision_number":5},{"id":"25a2e34ea875ce1f70b5281e4b88334dd36e6249","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 23:47:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10f6b09a9fbb438db41e395fa2513bc5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/925a282b49d24925931e4cd9a10800e8 : SUCCESS in 21m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/875bdc665ce34d12bf78bf6498930321 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6baf1388cb2f47d4949351cf835ab17e : RETRY_LIMIT in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"36aa2cf5e8a2196b6eae915dc4f5c792ddb1e24b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-24 21:46:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-07-958707-5-fc : SUCCESS in 3h 3m 4s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-07-958707-5-iscsi : FAILURE in 1h 58m 23s","accounts_in_message":[],"_revision_number":5},{"id":"f7d343045668d5979fc6508a6ab41b280e6a4f2c","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-01-28 15:12:14.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cab279e84708c09e713bf2e50797f4f08a2e5264","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-04 15:39:14.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"93b7d5c13a80c44eada0470c206b394d6b1405d2","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-02-20 14:43:08.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b27e3096b0819a197daa554a6fdb56edf13681bd","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-20 16:58:29.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a4a2ca433d286b7ac8d6e9ce42ee09f9743aa3ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2026-04-02 19:43:48.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fe220914e1bbfac365e01c1b43807bc4e1847cf2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-02 19:50:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/85462c8ecbc64499829db6550950e2ad : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/aa686a77ef8443bdbaa716411b8ec229 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":6},{"id":"18e4a44cb52cbde19857278d1853eba267c5682b","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2026-04-02 19:53:51.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7e915cec7710f69c163cc6d1ae4857ff75dd4787","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2026-04-02 19:54:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"afbf5a38ae280bc90fddd6428c069642ca49e8db","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-02 20:00:58.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/506aa482df10457381bce9b7f0f2987e : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8472fc9bb762424a9d7545647907c7e2 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":7},{"id":"34c98b61b0f4ce597db55bfb7c767fa9eb05524e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 20:12:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98cbf3bbc55b40f78803c43fd28cd3ff\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f48260706a0439c85ea1f7b805b95e5 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0424cd7486ff4cdc81ebc631aef58086 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bd44b560bc9f43e2897653ac20087935 : SUCCESS in 15m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8c8dc1466a1c91acf0e02e541787d9c87c3ad568","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-02 20:22:17.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c .","accounts_in_message":[],"_revision_number":6},{"id":"430f8a6892b0a284122beb114eb66cf509acf215","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-02 20:32:21.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c .","accounts_in_message":[],"_revision_number":7},{"id":"722239c8e6326499391a3e2c048777e184c30137","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-02 21:03:12.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/ecce97cc9a554d018f927586d004a9c9 : FAILURE in 40m 57s","accounts_in_message":[],"_revision_number":7},{"id":"fe3728ffbdbec2f79e0c0b2b7f3b3b607f16fce5","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-02 21:50:16.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver build/0de56d0e16ae4e758cadf157958b8af2 : FAILURE in 31m 09s","accounts_in_message":[],"_revision_number":7},{"id":"9a60c01afc0268869a08c2f6ff4991658a505d59","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-02 22:03:05.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958707 : FAILURE in 42m 17s","accounts_in_message":[],"_revision_number":7},{"id":"6185d9807bcab0866285c16d6da541d2895bf939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 22:04:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1993c5336779429b82af5df99f8f2cfb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c8120cdf80b4fd99f892117a20fd67d : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c666dc066c8d4589a2fb6253626dbeae : SUCCESS in 8m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba50bb1f784c495fb4abe7760559b418 : SUCCESS in 6m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2e7b603dc7ed4b83a7177a6f307835fe : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fddce6d9f4545d5b6a9ec81889ba22f : SUCCESS in 7m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/d0d927a621c8435091973e67d5bf3a14 : SUCCESS in 58m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ccf7248db7d490eb5000601716e516b : SUCCESS in 29m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f36eed01c0604eefa48ec96607cdfbaf : SUCCESS in 1h 21m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/408c1bb03b674728aff8cbafa4a014fc : SUCCESS in 36m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ba8dd44494f432cb3e3811c988eb244 : SUCCESS in 13m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eb82825895a1435fa0b7389a38f1e7f3 : SUCCESS in 16m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ca39a59aa04847978691722e66fb0b3e : SUCCESS in 2m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/50ba00e97fbc4df3bb60e2d07a963a68 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b90b1aec48e640ee826c49a23099b0ea : SUCCESS in 5m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a564f98b942244d29560790cd7136118 : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/baf16f2a7d884689884b1ee573118ba8 : FAILURE in 21m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9fea0489f39d49b984b16204305dc0e0 : SUCCESS in 3m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a352cda813445569ed677d44113ffe7 : SUCCESS in 1h 28m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d173d0183a164df4a2bfbb2de079b636 : FAILURE in 2h 06m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b06a2c08c42941999fa3c092146cd9dd : SUCCESS in 1h 55m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1eacc0c67fbb4f71ac2ff3e8cac03eea : SUCCESS in 1h 38m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3999303f7deb4773b476cc38499b8692 : SUCCESS in 23m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c32eecdd4ab49c0bfaeed7864347420 : SUCCESS in 1h 12m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/482097970d2d4fffa4b41a8dfea9c293 : SUCCESS in 1h 31m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/477ec95cde9545c3acf95916031cf8f1 : SUCCESS in 45m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/87cbf4fc05c54102aebd558e6c636923 : SUCCESS in 1h 57m 01s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2c69d72aba914ee5954848acd3b245ab : SUCCESS in 1h 46m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/53919f77efec4be1b6b97e3e396e614c : SUCCESS in 1h 57m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f04b5aeea4a43ea88b837f108133d1f : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":7},{"id":"8ca75b482c9398e941260edd8e01d4db96e549a7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-02 22:13:38.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_04_2026_22_07_17\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"64ceec189f5624e9ea2281aa13e7e77e3b17f398","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-02 23:09:05.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/717d9bcf57fa4a76aa44ab843930c5b6 : FAILURE in 40m 11s","accounts_in_message":[],"_revision_number":7},{"id":"086ab9a84b9eec0bf2daed3dcd96a6d25d0413eb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-03 08:21:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3694/.","accounts_in_message":[],"_revision_number":6},{"id":"06644d53dbf91169ffcc19caa3abffe5a4f49e2e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-03 09:42:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3695/.","accounts_in_message":[],"_revision_number":7},{"id":"59da12b232ebfe23dd9c413fc93faf7e5decc526","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-06 09:45:11.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3518/.","accounts_in_message":[],"_revision_number":6},{"id":"15d72546ede4c858a0e6d91a5d6a6eaef8c4b5dd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-06 09:58:58.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3519/.","accounts_in_message":[],"_revision_number":7},{"id":"75e33557342bead2bf707905b07e5581cb1b09f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:50:05.000000000","message":"Patch Set 7:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/1346/ : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"f1706f2a71cef9277de7c12498623007b0c6dfd0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:56:01.000000000","message":"Patch Set 6:\n\nBuild Failed","accounts_in_message":[],"_revision_number":6}],"current_revision_number":7,"current_revision":"b821a68baed1517bd102378e2aa8504c76f483e3","revisions":{"0683f4eecc4769a26bd801cd1cb1ede65d1d2c93":{"kind":"REWORK","_number":1,"created":"2025-08-27 22:35:49.000000000","uploader":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"ref":"refs/changes/07/958707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/958707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/958707/1"}}},"commit":{"parents":[{"commit":"89709ef589b74506885c72e42cad57eb8dedfb56","subject":"Merge \"mypy: Bump to 1.17.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89709ef589b74506885c72e42cad57eb8dedfb56"}]}],"author":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2025-08-19 17:40:15.000000000","tz":-180},"committer":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2025-08-27 20:35:17.000000000","tz":-180},"subject":"Improve restoration of encrypted volume backups","message":"Improve restoration of encrypted volume backups\n\nWhen restoring a backup of an encrypted volume, Cinder currently\nrequires the destination volume type ID to be identical to the source\nvolume type ID [1]. This serves as a safeguard to ensure the encryption\ncontexts match.\n\nHowever, this reliance on matching opaque IDs creates a limitation.\nSince volume type IDs are unique, it prevents restoring a backup to\na newly created volume type, even if its encryption settings are\nidentical to the original.\n\nWhile the concern behind this requirement is valid, a better approach is\nto verify compatibility by comparing the encryption settings directly.\n\nTo enable this, this change stores the source volume\u0027s encryption\nmetadata within the backup itself. During a restore, this stored\ninformation is used to ensure the target volume type has compatible\nencryption settings, allowing the process to proceed securely.\n\nBackups created prior to this update will restore using the legacy logic\nas they lack this encryption metadata. New backups will only restore if\nthe source and destination encryption configurations are identical.\n\n[1] https://opendev.org/openstack/cinder/src/commit/46cc51014226422b84d08f63a1d97dd13639d89b/cinder/backup/driver.py#L213\n\nCloses-Bug: #2121557\nChange-Id: Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c\nSigned-off-by: Luan Utimura \u003cluan.utimura@luizalabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0683f4eecc4769a26bd801cd1cb1ede65d1d2c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0683f4eecc4769a26bd801cd1cb1ede65d1d2c93"}]},"branch":"refs/heads/master"},"b370b0d1dbf793b0c32de54e531bbb24cdf2ee11":{"kind":"REWORK","_number":2,"created":"2025-08-28 12:34:30.000000000","uploader":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"ref":"refs/changes/07/958707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/958707/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/958707/2"}}},"commit":{"parents":[{"commit":"89709ef589b74506885c72e42cad57eb8dedfb56","subject":"Merge \"mypy: Bump to 1.17.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89709ef589b74506885c72e42cad57eb8dedfb56"}]}],"author":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2025-08-19 17:40:15.000000000","tz":-180},"committer":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2025-08-28 12:30:54.000000000","tz":-180},"subject":"Improve restoration of encrypted volume backups","message":"Improve restoration of encrypted volume backups\n\nWhen restoring a backup of an encrypted volume, Cinder currently\nrequires the destination volume type ID to be identical to the source\nvolume type ID [1]. This serves as a safeguard to ensure the encryption\ncontexts match.\n\nHowever, this reliance on matching opaque IDs creates a limitation.\nSince volume type IDs are unique, it prevents restoring a backup to\na newly created volume type, even if its encryption settings are\nidentical to the original.\n\nWhile the concern behind this requirement is valid, a better approach is\nto verify compatibility by comparing the encryption settings directly.\n\nTo enable this, this change stores the source volume\u0027s encryption\nmetadata within the backup itself. During a restore, this stored\ninformation is used to ensure the target volume type has compatible\nencryption settings, allowing the process to proceed securely.\n\nBackups created prior to this update will restore using the legacy logic\nas they lack this encryption metadata. New backups will only restore if\nthe source and destination encryption configurations are identical.\n\n[1] https://opendev.org/openstack/cinder/src/commit/46cc51014226422b84d08f63a1d97dd13639d89b/cinder/backup/driver.py#L213\n\nCloses-Bug: #2121557\nChange-Id: Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c\nSigned-off-by: Luan Utimura \u003cluan.utimura@luizalabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b370b0d1dbf793b0c32de54e531bbb24cdf2ee11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b370b0d1dbf793b0c32de54e531bbb24cdf2ee11"}]},"branch":"refs/heads/master"},"cb14b0d456496123dcfeb2bd7d4597fb221a892a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-08-28 20:19:32.000000000","uploader":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"ref":"refs/changes/07/958707/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/958707/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/958707/3"}}},"commit":{"parents":[{"commit":"55d27f77a6e699738b8f96889d60b931a9fc4a9e","subject":"Merge \"Added support for NetApp ASA r2 (All-Flash SAN Array r2) disaggregated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55d27f77a6e699738b8f96889d60b931a9fc4a9e"}]}],"author":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2025-08-19 17:40:15.000000000","tz":-180},"committer":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2025-08-28 20:10:26.000000000","tz":-180},"subject":"Improve restoration of encrypted volume backups","message":"Improve restoration of encrypted volume backups\n\nWhen restoring a backup of an encrypted volume, Cinder currently\nrequires the destination volume type ID to be identical to the source\nvolume type ID [1]. This serves as a safeguard to ensure the encryption\ncontexts match.\n\nHowever, this reliance on matching opaque IDs creates a limitation.\nSince volume type IDs are unique, it prevents restoring a backup to\na newly created volume type, even if its encryption settings are\nidentical to the original.\n\nWhile the concern behind this requirement is valid, a better approach is\nto verify compatibility by comparing the encryption settings directly.\n\nTo enable this, this change stores the source volume\u0027s encryption\nmetadata within the backup itself. During a restore, this stored\ninformation is used to ensure the target volume type has compatible\nencryption settings, allowing the process to proceed securely.\n\nBackups created prior to this update will restore using the legacy logic\nas they lack this encryption metadata. New backups will only restore if\nthe source and destination encryption configurations are identical.\n\n[1] https://opendev.org/openstack/cinder/src/commit/46cc51014226422b84d08f63a1d97dd13639d89b/cinder/backup/driver.py#L213\n\nCloses-Bug: #2121557\nChange-Id: Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c\nSigned-off-by: Luan Utimura \u003cluan.utimura@luizalabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb14b0d456496123dcfeb2bd7d4597fb221a892a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb14b0d456496123dcfeb2bd7d4597fb221a892a"}]},"branch":"refs/heads/master"},"38ae9fa184a9d1f128a1536f0d0927fc391025a0":{"kind":"REWORK","_number":4,"created":"2025-11-25 13:32:50.000000000","uploader":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"ref":"refs/changes/07/958707/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/958707/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/958707/4"}}},"commit":{"parents":[{"commit":"5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23","subject":"Merge \"Hitachi: support extending volume has snapshot(s)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23"}]}],"author":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2025-08-19 17:40:15.000000000","tz":-180},"committer":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2025-11-25 13:32:21.000000000","tz":-180},"subject":"Improve restoration of encrypted volume backups","message":"Improve restoration of encrypted volume backups\n\nWhen restoring a backup of an encrypted volume, Cinder currently\nrequires the destination volume type ID to be identical to the source\nvolume type ID [1]. This serves as a safeguard to ensure the encryption\ncontexts match.\n\nHowever, this reliance on matching opaque IDs creates a limitation.\nSince volume type IDs are unique, it prevents restoring a backup to\na newly created volume type, even if its encryption settings are\nidentical to the original.\n\nWhile the concern behind this requirement is valid, a better approach is\nto verify compatibility by comparing the encryption settings directly.\n\nTo enable this, this change stores the source volume\u0027s encryption\nmetadata within the backup itself. During a restore, this stored\ninformation is used to ensure the target volume type has compatible\nencryption settings, allowing the process to proceed securely.\n\nBackups created prior to this update will restore using the legacy logic\nas they lack this encryption metadata. New backups will only restore if\nthe source and destination encryption configurations are identical.\n\n[1] https://opendev.org/openstack/cinder/src/commit/46cc51014226422b84d08f63a1d97dd13639d89b/cinder/backup/driver.py#L213\n\nCloses-Bug: #2121557\nChange-Id: Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c\nSigned-off-by: Luan Utimura \u003cluan.utimura@luizalabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38ae9fa184a9d1f128a1536f0d0927fc391025a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38ae9fa184a9d1f128a1536f0d0927fc391025a0"}]},"branch":"refs/heads/master"},"5e0e112fb2638d084c225838b2bb268687c9f811":{"kind":"REWORK","_number":5,"created":"2026-01-23 12:34:53.000000000","uploader":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"ref":"refs/changes/07/958707/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/958707/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/958707/5"}}},"commit":{"parents":[{"commit":"d583e34f8b0301f6f3dc4ec362006d8754ed21b0","subject":"Merge \"NetApp - Bug Fix 2124264: Pools not getting populated for ASAr2 platform\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d583e34f8b0301f6f3dc4ec362006d8754ed21b0"}]}],"author":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2025-08-19 17:40:15.000000000","tz":-180},"committer":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2026-01-23 12:34:36.000000000","tz":-180},"subject":"Improve restoration of encrypted volume backups","message":"Improve restoration of encrypted volume backups\n\nWhen restoring a backup of an encrypted volume, Cinder currently\nrequires the destination volume type ID to be identical to the source\nvolume type ID [1]. This serves as a safeguard to ensure the encryption\ncontexts match.\n\nHowever, this reliance on matching opaque IDs creates a limitation.\nSince volume type IDs are unique, it prevents restoring a backup to\na newly created volume type, even if its encryption settings are\nidentical to the original.\n\nWhile the concern behind this requirement is valid, a better approach is\nto verify compatibility by comparing the encryption settings directly.\n\nTo enable this, this change stores the source volume\u0027s encryption\nmetadata within the backup itself. During a restore, this stored\ninformation is used to ensure the target volume type has compatible\nencryption settings, allowing the process to proceed securely.\n\nBackups created prior to this update will restore using the legacy logic\nas they lack this encryption metadata. New backups will only restore if\nthe source and destination encryption configurations are identical.\n\n[1] https://opendev.org/openstack/cinder/src/commit/46cc51014226422b84d08f63a1d97dd13639d89b/cinder/backup/driver.py#L213\n\nCloses-Bug: #2121557\nChange-Id: Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c\nSigned-off-by: Luan Utimura \u003cluan.utimura@luizalabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e0e112fb2638d084c225838b2bb268687c9f811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e0e112fb2638d084c225838b2bb268687c9f811"}]},"branch":"refs/heads/master"},"ae0a5dabe88ba1058a4746c089c37ed01965af14":{"kind":"REWORK","_number":6,"created":"2026-04-02 19:43:48.000000000","uploader":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"ref":"refs/changes/07/958707/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/958707/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/958707/6"}}},"commit":{"parents":[{"commit":"81c48732b13b4512ca63ac0fc91ccd7b0eec2863","subject":"Merge \"Cleanup glance image metadata during instance reschedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c48732b13b4512ca63ac0fc91ccd7b0eec2863"}]}],"author":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2025-08-19 17:40:15.000000000","tz":-180},"committer":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2026-04-02 19:43:32.000000000","tz":-180},"subject":"Improve restoration of encrypted volume backups","message":"Improve restoration of encrypted volume backups\n\nWhen restoring a backup of an encrypted volume, Cinder currently\nrequires the destination volume type ID to be identical to the source\nvolume type ID [1]. This serves as a safeguard to ensure the encryption\ncontexts match.\n\nHowever, this reliance on matching opaque IDs creates a limitation.\nSince volume type IDs are unique, it prevents restoring a backup to\na newly created volume type, even if its encryption settings are\nidentical to the original.\n\nWhile the concern behind this requirement is valid, a better approach is\nto verify compatibility by comparing the encryption settings directly.\n\nTo enable this, this change stores the source volume\u0027s encryption\nmetadata within the backup itself. During a restore, this stored\ninformation is used to ensure the target volume type has compatible\nencryption settings, allowing the process to proceed securely.\n\nBackups created prior to this update will restore using the legacy logic\nas they lack this encryption metadata. New backups will only restore if\nthe source and destination encryption configurations are identical.\n\n[1] https://opendev.org/openstack/cinder/src/commit/46cc51014226422b84d08f63a1d97dd13639d89b/cinder/backup/driver.py#L213\n\nCloses-Bug: #2121557\nChange-Id: Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c\nSigned-off-by: Luan Utimura \u003cluan.utimura@luizalabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae0a5dabe88ba1058a4746c089c37ed01965af14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae0a5dabe88ba1058a4746c089c37ed01965af14"}]},"branch":"refs/heads/master"},"b821a68baed1517bd102378e2aa8504c76f483e3":{"kind":"REWORK","_number":7,"created":"2026-04-02 19:54:50.000000000","uploader":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"ref":"refs/changes/07/958707/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/958707/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/958707/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/958707/7"}}},"commit":{"parents":[{"commit":"81c48732b13b4512ca63ac0fc91ccd7b0eec2863","subject":"Merge \"Cleanup glance image metadata during instance reschedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c48732b13b4512ca63ac0fc91ccd7b0eec2863"}]}],"author":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2025-08-19 17:40:15.000000000","tz":-180},"committer":{"name":"Luan Utimura","email":"luan.utimura@luizalabs.com","date":"2026-04-02 19:54:41.000000000","tz":-180},"subject":"Improve restoration of encrypted volume backups","message":"Improve restoration of encrypted volume backups\n\nWhen restoring a backup of an encrypted volume, Cinder currently\nrequires the destination volume type ID to be identical to the source\nvolume type ID [1]. This serves as a safeguard to ensure the encryption\ncontexts match.\n\nHowever, this reliance on matching opaque IDs creates a limitation.\nSince volume type IDs are unique, it prevents restoring a backup to\na newly created volume type, even if its encryption settings are\nidentical to the original.\n\nWhile the concern behind this requirement is valid, a better approach is\nto verify compatibility by comparing the encryption settings directly.\n\nTo enable this, this change stores the source volume\u0027s encryption\nmetadata within the backup itself. During a restore, this stored\ninformation is used to ensure the target volume type has compatible\nencryption settings, allowing the process to proceed securely.\n\nBackups created prior to this update will restore using the legacy logic\nas they lack this encryption metadata. New backups will only restore if\nthe source and destination encryption configurations are identical.\n\n[1] https://opendev.org/openstack/cinder/src/commit/46cc51014226422b84d08f63a1d97dd13639d89b/cinder/backup/driver.py#L213\n\nCloses-Bug: #2121557\nChange-Id: Ibdd2b88db04135f69e4eddf30bd3a6700f9ec40c\nSigned-off-by: Luan Utimura \u003cluan.utimura@luizalabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b821a68baed1517bd102378e2aa8504c76f483e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b821a68baed1517bd102378e2aa8504c76f483e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
