)]}'
{"id":"openstack%2Fcinder~958938","triplet_id":"openstack%2Fcinder~master~I7ab8f7b57e13079f3d35354691c02e1f1ad930f4","project":"openstack/cinder","branch":"master","topic":"powermax-nvme-tcp-followup","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-09-01 10:48:30.000000000","reason":"Change was submitted"},"36725":{"account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"last_update":"2025-09-01 10:48:30.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-09-01 08:47:57.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I7ab8f7b57e13079f3d35354691c02e1f1ad930f4","subject":"Dell PowerMax: NVMe/TCP suuport (follow-up)","status":"MERGED","created":"2025-08-30 04:57:23.000000000","updated":"2025-09-01 10:53:57.000000000","submitted":"2025-09-01 10:48:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"958938-powermax-nvme-tcp-followup","meta_rev_id":"87ab00b2f2b5c3b42b6e8b0a2b8488902d098420","_number":958938,"virtual_id_number":958938,"owner":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-01 10:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-01 08:47:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-01 08:47:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-30 04:57:32.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-08-30 04:58:51.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-08-30 05:56:26.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-30 06:09:17.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-30 06:43:51.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-08-30 06:46:46.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-30 07:38:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-30 08:48:00.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-08-30 22:48:20.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"},{"updated":"2025-08-31 04:25:11.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-08-31 04:25:11.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"f666f60e311b47aaef38b470dc74ba138802875f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-30 04:57:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4a22382291b25d3339a843ac070d7fc2a53ebc9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-30 04:57:32.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/369/","accounts_in_message":[],"_revision_number":1},{"id":"6c64fa50bae69cea26dd9483435ec86c4fa43fe4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-30 04:58:51.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"33b9f7ff760f3e3a695d7a76fc2dc4fc50dafc3b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-30 05:54:12.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958938/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958938/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"312fc007e8832e157b5d1b01757558d1198336c4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-30 05:56:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1625/.","accounts_in_message":[],"_revision_number":1},{"id":"ddbd0c2efd2ffc3d92fa984e00ab787a634ff0e7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-30 05:59:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1759/.","accounts_in_message":[],"_revision_number":1},{"id":"13318a24cefc0d7c124ac77a4e4a3c40547b09e2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-30 06:09:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9969d24fa22b4ec69b28ede8db5160d0 : SUCCESS in 58m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0cdb9809fed949e19082d4902a1e7106 : FAILURE in 1h 10m 18s","accounts_in_message":[],"_revision_number":1},{"id":"dfc8b2cc09c0cc3e071ce9e88ec6d2ea0e675206","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-30 06:43:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958938 : FAILURE in 1h 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e7e120018903397eee0a4eca20baf6a6d2600d55","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-30 06:46:46.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"0c4284323a2b9710c3267a2da99441f6fdaceda6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 07:38:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecea045ccaec4812939c31d1b4505f2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fad30297d7f64f5996c200e020410cd2 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/522da5e769fd4ef7a9681e4baa20a9b9 : SUCCESS in 7m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ce0d38c9d494751bc6cf3ed5291d5eb : SUCCESS in 8m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36867be8f10a48ce8dae5a07723ac7e2 : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0cdc652ffbd4decbeec655966489877 : SUCCESS in 7m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/8a969964b3664520a0a300eb3b5010ac : SUCCESS in 30m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/607bdb385dff4138bd3264fbb9a08c6b : SUCCESS in 47m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1ecb2cfcc2b04f86a441db72743f42f6 : SUCCESS in 1h 26m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd3fcb401b534543a136414a77c42a00 : SUCCESS in 58m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fc8b147cc10845c0a839205d7fc55d60 : TIMED_OUT in 40m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/448bb3c054864a35b0323500b6ef3b9d : SUCCESS in 4m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7337d3e2164c43f79dbb4a48ebd27d44 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d1e9e1b4f3514fae8b9cc662ee4f3fff : SUCCESS in 5m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8c537c06f0c7488080040c3de86efd06 : SUCCESS in 11m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cf6be589bb234572bf3aa36bbe0da9a8 : FAILURE in 17m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4391779fc6f74f878cae9b5d7bc5e536 : SUCCESS in 4m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fcf7d1ea82904949b14bbf0dcb59e605 : SUCCESS in 1h 54m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/31783fdae7724e3189e5ce65fecb0977 : FAILURE in 2h 35m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1579c2951eb0434cb28887fe7118f19e : SUCCESS in 1h 51m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1c93581ceb054b4298c117b7827c0434 : FAILURE in 10m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1a240604576f479ab11938fa1039b557 : SUCCESS in 25m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/13c28ba994f34288b19db2838ab7fbeb : SUCCESS in 1h 06m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/347ccf95de9a45eeaddcb6f89415066d : SUCCESS in 1h 31m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/51f39d8bce144e3aaa92fa4d919c914e : SUCCESS in 24m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d93ebf13529e4d77b2d31fbf2772e0af : SUCCESS in 1h 30m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6565d57384234089b5527d5ebc83f97b : FAILURE in 18m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86c5d9c4cdec4da1bd82db4e773630ab : SUCCESS in 1h 33m 03s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f88ddc655a24457e96e217edf8c1dd63 : SUCCESS in 1h 16m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edceb8aec1704d8fa8d0542b89cfb78f : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e7e3cb8c4ce276dfceea52a1c3fe09cd616d9351","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 08:24:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e3e9be1b97c4ba7960fc0fe9418af94\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/93bc74e1ed9e425c9f30cc5c5ef49bbe : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b95541a129b6416197f39d0bc3dc8889 : SUCCESS in 26m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a59d1179dc2d48b49474d83c85786a4d : SUCCESS in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d718b7d893a9c43c9cbf9f933206beaccb8e7e68","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-30 08:48:00.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-38-958938-1/FC/ : UNSTABLE in 14m 22s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-38-958938-1/iSCSI/ : FAILURE in 3h 48m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"2cafa8fc4ed3b41bac265e904fe85bfd44dd9a38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-30 08:58:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cd4a57d09a4a6dbb857273b38a9c1aafda0a39e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-30 08:58:18.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/370/","accounts_in_message":[],"_revision_number":2},{"id":"6ce77885b2c60d6c0ea7757d8e1e7ac329874f57","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-08-30 08:59:32.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"5afd963eec68c4f47be0b53af579491cb9842c65","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-30 09:02:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"87cf791df25163736b781dbf3b27540c6ef45002","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 09:28:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1af892e1237b4b299da3d95e0bc29f5f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/05f2761185c34f3eb9b5cc0e8301fe6d : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d62c54cc643741cb852319ea39f82cb7 : SUCCESS in 22m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0e837c9eb14e464f8fcf8dff6d3017d9 : SUCCESS in 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e885078a41945fc25525d628ba3c63b7147bd63","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-30 09:55:45.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958938/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/958938/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"7b2bbbc1e5bfe9d63dea96bde8928cdfba500905","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-30 09:59:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1626/.","accounts_in_message":[],"_revision_number":2},{"id":"e434a3210b7e9e4231c1a473ab4b02fecae2a6fb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-30 10:01:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1760/.","accounts_in_message":[],"_revision_number":2},{"id":"304b3a7bb854391d2795aee066572328e7203aa5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-30 10:04:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f3825654af3845238b9764c3ba927d03 : SUCCESS in 47m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/907a920430be451abd648128046d5c9b : FAILURE in 1h 04m 51s","accounts_in_message":[],"_revision_number":2},{"id":"0c61f36ae40de12bf5ab743ea9599fc6842e40e4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-30 10:25:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/958938 : FAILURE in 1h 26m 00s","accounts_in_message":[],"_revision_number":2},{"id":"29884b2134062048a8df7692f18005c577b210aa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-08-30 10:43:20.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"bc7664c3b75b259ed15f43d8c63551e7f5a6b107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 10:45:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22c224cfb4e847458acc3ec7ae42fd69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963c74ac05894c1597f97f6027d6ac46 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d05ac35987b04eb498b301fcccd1045e : SUCCESS in 7m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ad8cb71d42f491789bfdf72c694945c : SUCCESS in 7m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d28c1c8b4dc44caea8af23063ef90dac : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/937832c6c0d841be95ac907ba9656b61 : SUCCESS in 7m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/10c65ef18da54ce0bb9adb7371ffd10d : SUCCESS in 31m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cbb47fa10cbd47a19f700593a4235303 : SUCCESS in 32m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9621f3ee3d8f4dbcbcd7179068c556c0 : SUCCESS in 1h 20m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30d155daa2594258b60decc8e424de48 : SUCCESS in 1h 02m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/106956e9d92746708dfe5b835fc53aae : SUCCESS in 16m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1a2dcf426b114252a078d02b980ae4e6 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e6883fe625744e5184164d8648fae6de : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c3440d130cf4465b11756d7b4170ece : SUCCESS in 5m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f74599621fa64f52961eb20cac3cc84b : SUCCESS in 8m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d6f3e5e2b75c4cd484ffda50df932183 : FAILURE in 35m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/133bfd647c924b139a752772b83a8a0c : SUCCESS in 7m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/83113cafa1cd49e88d8afedaf07c6318 : FAILURE in 1h 01m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c493f4d7d23f4107a6a18b4c677228fb : FAILURE in 1h 36m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e2b14cfda64345bd94d8126ff19f1f64 : SUCCESS in 1h 45m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a9788869c3dd473bbe906051f22e889f : FAILURE in 8m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e28146115bed4348bf7366832c5f37f6 : SUCCESS in 18m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3aca15e31af34f82bd21b4dd0556c8eb : SUCCESS in 36m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5c7acf61bfb2450495a6870eab69a3fc : SUCCESS in 1h 19m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9a8a786481d4446091ed3b51b87b3dda : SUCCESS in 24m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/28ed58ace07849d59cffd041507bff6d : SUCCESS in 46m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/344006d518664afe985a99fedce29d9e : FAILURE in 10m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/726a8de92db8458dadfa9773a38ff293 : SUCCESS in 1h 15m 01s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d337d93576ef4a9ba700dc4b5f3f1d1d : SUCCESS in 1h 28m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4d67429f0cb40539655206a5c37dd73 : SUCCESS in 30m 58s","accounts_in_message":[],"_revision_number":2},{"id":"f5e49f977057c113b066b066f433fd7dbebcee9c","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-30 11:08:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb536e869ea228215e61593789cbdc9cf95eb0e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 11:30:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d485f4c6cca4a89ae77624f9eab6fca\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc9b606031fa42478ab3763447503092 : SUCCESS in 18m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/381d0fb30a8648c38e533feda41f4e99 : SUCCESS in 19m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7d148f408d02442786b0c48b30a21455 : SUCCESS in 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8c70f69972940d1009ee5a7eba0dc44e91d1969","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-30 12:07:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/36ff5913d6694ea4967977fc0b9194ad : SUCCESS in 51m 40s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b4c19420d5a945d2a42ffa15caddecea : FAILURE in 58m 14s","accounts_in_message":[],"_revision_number":2},{"id":"1f4df22220bea58184f633c362073b24e0fc892f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 14:18:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/775f30e53cf44bfcb25f283b5609b347\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d5f4f1c731a4adf9094900d7c4bd553 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec9976092085484780195732a273cad9 : SUCCESS in 7m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/606d7fb2dbe343cc9c1a1ae0868914b6 : SUCCESS in 11m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1828f4f8182543b6b5c9594873022b2d : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d179344acb554d30bde16d2973e40a30 : SUCCESS in 6m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/677c5cb7bcd04b2bbdaaf3deee387d28 : SUCCESS in 27m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd8cd208014d4913ae4aeb2692b8a622 : SUCCESS in 35m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32a7fbeb4894459286611e0161e30ff5 : SUCCESS in 1h 07m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/404555eeb9ce4321be178ecb9d8797aa : POST_FAILURE in 36m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c68cbf84f9fd4e9c9f0e86da61764f5c : TIMED_OUT in 40m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c5568eabd4a44be3830e586c15945f3a : SUCCESS in 4m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/530c8be69b9e4ee19ecc10912ef01918 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9bf8c593eaa241e7b3684537b5a63c51 : SUCCESS in 7m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d744936c7a9d4dfeb619114571ed8d83 : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cdf6778fa0d84922b4674ac1a19ad319 : FAILURE in 36m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2e4cf48fd2dc48d7a859f16accc1e110 : SUCCESS in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f267d90a58e84afcbde604692696fd37 : SUCCESS in 1h 29m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/79af8cea1df3453180487dd6932c6725 : FAILURE in 3h 04m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/061d59c632e447b0bd0d6a6297193287 : SUCCESS in 1h 18m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0980173731264f4b8bb882de03ae0d85 : FAILURE in 15m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a62f8a3aa78a4c78949a1a63fd8f69a2 : SUCCESS in 34m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/18b1881879e643509542e456894bb0b8 : SUCCESS in 1h 05m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/defac2c189244999bafcff3468fd3faa : SUCCESS in 1h 27m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/691730370ae4419a85fa91b9ae6e6b3a : SUCCESS in 42m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/709038eb39674322badafcdd682f142c : POST_FAILURE in 1h 48m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1effd6fb3b8b4be8942a3f3e36a41f7e : FAILURE in 8m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/965cade3ed1a431e91354537afb0c414 : SUCCESS in 50m 47s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/928ce7393d4548a9b41e14455f0c20aa : SUCCESS in 1h 22m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5af0e61ec1f4594a87c04c557fdf6ef : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":2},{"id":"38286daaa9cf668b21e452b0f837d6ff2e7f5857","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-30 16:47:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"38eccc09354e5ce6130e8c236c7d1932d33acedb","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-30 16:47:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22807d3023da86f3ed1c850f831eb725b7b73f25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 17:13:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f6dbc3fcb9c46a8b13535ab3cec3e29\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be9d85b8f74e4acfaa81c614068c704c : SUCCESS in 16m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab7e31f0df374efea9020acdb42d8ac3 : SUCCESS in 19m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/983251dbe92d42439a9608a76b724703 : SUCCESS in 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e53bee6504641cdece8361958f3aeeab41f0f83","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-30 17:42:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/639f518f5958463e9b3f4890cf429268 : SUCCESS in 52m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8d58f6ee4fa741eb9cb8b1649fe7639a : FAILURE in 50m 46s","accounts_in_message":[],"_revision_number":2},{"id":"19e78609b60c1d89bd59eb112e76365d962081f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 19:33:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46843c76183f417c8dc11a620c3c0187\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d62a30ba8f549ccbc8569a503109437 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41cec94231bd427881322265e90417d4 : SUCCESS in 9m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f3dcde8922c4ab487f4fd0fa7a3a019 : SUCCESS in 11m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61937d07bbe64eada9101a984708eb7b : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd2f39ba88541848880529bb84f86f9 : SUCCESS in 13m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/ce3f5fa6fba5441cb822b3a32c4390b9 : SUCCESS in 1h 00m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f6a5a42eddd43d5bd46ef0ee5a3d9d7 : SUCCESS in 1h 00m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/06a564a788d340c3863ca7fd9beaf1b7 : SUCCESS in 1h 41m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e416fc65c1784ef48ae5b1c73b6c7217 : SUCCESS in 43m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0340fc55d78744dbaaa70fb82239d41c : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/410838c6dc6b46df88eaba026828eb1a : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f4e5069f57c4ba6aab401b8ced55bc2 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/498eb701cf8d4f4f99836e34c36f8643 : SUCCESS in 4m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dba414b1ee0c4c58a15dc0f89aaa2f3c : SUCCESS in 8m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e1cacfd8c0fc407092e28e0dd5df22db : FAILURE in 17m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b052c52ba8614220b7be29ecbf407113 : SUCCESS in 7m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/04edb447da7445839f7c3a5621e14953 : SUCCESS in 1h 22m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f3c37508d8aa4e3d9341c75cf7205341 : FAILURE in 2h 32m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f31008fdb25041579beb9caf271bc956 : SUCCESS in 1h 57m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/77e5ce1839e14be8b29d5e6c13e9a28a : FAILURE in 10m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c7e98f05c8474456b81ed452970e8fcc : SUCCESS in 34m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/40af8f6686824eb199d24591efbf1688 : SUCCESS in 38m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/094e91d46e4843f1bb38149abf53a445 : SUCCESS in 1h 25m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/42301c13eb594e25a22becb18fca9240 : SUCCESS in 22m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/38f4ffa0e0bb4f64b4ab68f736c642d3 : SUCCESS in 1h 25m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ea16445e97f64ce8aec836524a5c6508 : FAILURE in 9m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a0a4d55ddda4b9eb7354b3c2d7741ed : SUCCESS in 1h 32m 25s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a68137d44f1e4ed9bf93b728d061e67e : SUCCESS in 1h 24m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb24d3d4c2bf480780a8f019ae892200 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":2},{"id":"4110dc304b42fe57c1783937f613a32279eb5022","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-30 22:03:11.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-38-958938-2/FC/ : FAILURE in 13h 2m 59s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-38-958938-2/iSCSI/ : FAILURE in 4h 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"31f5c47b62df5e2aacca05b119f0c9112faf4391","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2025-08-30 22:48:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\nhttps://elab-os-logsrv.delllabs.net/t/openstack/buildset/15e5e250a36242b09dad4d11bc86c8f4\n\n- powermax-iscsi-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/ef71cc05c79f4c2eb5141f6ea8fe7e91 : SUCCESS in 1h 56m 22s\n- powermax-fc-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/fc700554ed3c4679b597a1f881525619 : SUCCESS in 2h 09m 05s\n- powermax-nvme-tcp-cinder-tempest https://elab-os-logsrv.delllabs.net/t/openstack/build/90a25b17fd034548b7b0d18f88806c89 : SUCCESS in 1h 53m 03s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"f4171c1fce2a26bf0cdf890fae4ff3003c6e3254","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2025-08-31 04:25:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4cbaa9dafb0ae0189d169bc4496ccbc0406e0f11","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-09-01 08:47:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"df765b91a1e48391a2d5e591ac85b4018377875e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 08:48:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dd6e3a3485ac4828dc0de3d83d0c2baadce8b319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 10:48:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/add4ced500634fe496eebc506ab33dcc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969504a6504642428ea30a978b012e38 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e440579e88ee4085a97a23c64c2359bc : SUCCESS in 9m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d50e8478c4754cb1a488cea6fc8f98a9 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a9f28124292442db11b63a72fa2dd1b : SUCCESS in 7m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/f57bd752cee44876ab3ba63300b1c029 : SUCCESS in 1h 00m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92e920100bf54824b0118afd7b52fd7d : SUCCESS in 58m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c7be722d26254d7e93cda11efa92fec8 : SUCCESS in 1h 40m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d0eb65030514734b3ddac50d9d9ca28 : SUCCESS in 55m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0ce25f77cc364770a6128e2830e03400 : SUCCESS in 1h 11m 15s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2be1793d01194903b8cc0e2323db4110 : SUCCESS in 1h 56m 11s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/015e81ba0a594e6da1fabde3d885b73a : SUCCESS in 1h 26m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/501e0cab08f2471794cc654689676a4a : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a79814d52c1486b94cb193f74c7c6f26af869e46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 10:48:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"87ab00b2f2b5c3b42b6e8b0a2b8488902d098420","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 10:53:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c131b36fd0c44f2b115c24c7652c10e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d59f16370148908337f771e0f16dfd : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7926029d07f7de68f68c7468a85c48aedc238b1f","revisions":{"101982cc8d2412933635558f37aed0f02075b4ef":{"kind":"REWORK","_number":1,"created":"2025-08-30 04:57:23.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/38/958938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/958938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/958938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/958938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/958938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/958938/1"}}},"commit":{"parents":[{"commit":"4afd6cc37779a67ced50903c696353bd3ec1ade4","subject":"Merge \"NetApp: NVMe namespace mapping fails during VM live migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4afd6cc37779a67ced50903c696353bd3ec1ade4"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2025-08-30 04:55:29.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2025-08-30 04:57:01.000000000","tz":0},"subject":"Dell PowerMax: NVMe/TCP suuport (follow-up)","message":"Dell PowerMax: NVMe/TCP suuport (follow-up)\n\nChange-Id: I7ab8f7b57e13079f3d35354691c02e1f1ad930f4\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/101982cc8d2412933635558f37aed0f02075b4ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/101982cc8d2412933635558f37aed0f02075b4ef"}]},"branch":"refs/heads/master"},"7926029d07f7de68f68c7468a85c48aedc238b1f":{"kind":"REWORK","_number":2,"created":"2025-08-30 08:58:06.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/38/958938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/958938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/958938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/958938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/958938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/958938/2"}}},"commit":{"parents":[{"commit":"4afd6cc37779a67ced50903c696353bd3ec1ade4","subject":"Merge \"NetApp: NVMe namespace mapping fails during VM live migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4afd6cc37779a67ced50903c696353bd3ec1ade4"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2025-08-30 04:55:29.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2025-08-30 08:57:42.000000000","tz":0},"subject":"Dell PowerMax: NVMe/TCP suuport (follow-up)","message":"Dell PowerMax: NVMe/TCP suuport (follow-up)\n\nChange-Id: I7ab8f7b57e13079f3d35354691c02e1f1ad930f4\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7926029d07f7de68f68c7468a85c48aedc238b1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7926029d07f7de68f68c7468a85c48aedc238b1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
