)]}'
{"id":"openstack%2Fcinder~959001","triplet_id":"openstack%2Fcinder~master~Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa","project":"openstack/cinder","branch":"master","topic":"bp/ut-improvement-powerstore-driver","attention_set":{"38284":{"account":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"last_update":"2026-03-02 14:38:13.000000000","reason":"Someone else replied on the change"},"38286":{"account":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"last_update":"2026-05-05 12:06:52.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2026-01-09 07:01:08.000000000","reason":"removed on reply"},"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-09-08 12:58:03.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa","subject":"Dell Powerstore: Improve UT coverage","status":"NEW","created":"2025-09-01 06:10:47.000000000","updated":"2026-06-19 20:00:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a70855170d8f464df8f4515a08412bd5fcb04b75","_number":959001,"virtual_id_number":959001,"owner":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-18 14:04:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-01 06:50:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-01 06:59:09.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-09-01 06:59:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-01 07:59:50.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-01 08:09:57.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-09-01 08:10:08.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-01 09:01:15.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-09-01 10:32:45.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-09-02 08:14:54.000000000","updated_by":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"reviewer":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"state":"CC"},{"updated":"2025-09-02 12:35:50.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2025-09-08 10:46:39.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-08 18:02:48.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-11 05:54:21.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"CC"},{"updated":"2025-09-11 08:01:11.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-09-11 08:01:51.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-09-11 08:02:32.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-09-11 08:03:11.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2026-01-08 11:38:04.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-01-09 08:33:42.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-02-19 06:54:54.000000000","updated_by":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"reviewer":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"state":"CC"},{"updated":"2026-03-02 09:58:55.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-03-02 10:12:44.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-03-02 10:28:23.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-03-02 10:39:16.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-05 12:06:52.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-05 12:10:20.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-05 12:13:12.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-05 12:21:29.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-06-18 13:59:59.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"}],"messages":[{"id":"c49f9aa39ec34b8f96b2754b45101c99308dbfd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2025-09-01 06:10:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65e8597d35739dfd114f1697bbbee7984f8ee476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2025-09-01 06:18:03.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c0c1bd0358b7e59e9504f1153c1932f476ff1a8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 06:50:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/def38e10563e4636a960880447eaeccc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b26d9923accb46e3bcda2f718483ace8 : SUCCESS in 24m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/afc744f009c04b8da91b16c90ce8d9f7 : SUCCESS in 24m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e5a7dc4d69a042ec9bf09882106a9672 : SUCCESS in 30m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd7c749c69390dddb66567bc81b3f5b3f0bc3b7a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 06:59:09.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/379/","accounts_in_message":[],"_revision_number":1},{"id":"06ef27aab52374c7faa3c5de475477d9df2ecf63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 06:59:51.000000000","message":"Patch Set 2: Verified-1\n\n(258 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbc99c091e9f4f49abba3ea197869e4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c88aaf5e22ed45b293bc55aa8b465707 : FAILURE in 10m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9d18ac4b87442f2beea38d51e434884 : SUCCESS in 11m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4701e1176b594297ba596b2148b01044 : SUCCESS in 9m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3e7768411714755a95db3e7c3b6f7f1 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b558aa83dbc841d98fa3b57ed1fb7927 : SUCCESS in 9m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4df3f73ea3442779ab516cbdd90c0f4 : SUCCESS in 40m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5575a6f577c246e680abd1bae16ea0d7 : SUCCESS in 21m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ba74fa863ac4a1aaa5cc3d519fbe65c : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"790c596698b9c78df8063c286c849880414321ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 07:54:54.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"7574195426639c4c0fbd24cbb81f05fb528acb88","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 07:54:56.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/380/","accounts_in_message":[],"_revision_number":2},{"id":"a9aae06e5d1f9ecb077eae5756b20ad0ee399eca","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-01 07:59:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/37711dd71abc447cad94abb51d1ecb6c : SUCCESS in 48m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ab43a9c33f1e465e9e7cabeae8989b2b : FAILURE in 59m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5fe2cd4520f59004deea38f01edeb0601368666a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-01 08:09:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1767/.","accounts_in_message":[],"_revision_number":2},{"id":"0ca890580976fdf0de484773f2349411ccc5051a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-01 08:10:08.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"9a9379333ae84dec3d480284545cdf56fc5b2f1a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-01 08:12:58.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"554b3428355e1222c6c66d7753a77d017591fa5b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 08:51:15.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"f70cdff3c6f7f26a50dc1d80460ed4359b75722c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-01 08:51:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1633/.","accounts_in_message":[],"_revision_number":2},{"id":"19a1a6bc02cdfc38d72b0ba32025f55dca77af2f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-01 09:01:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959001 : FAILURE in 1h 25m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ceedc6fa07724c4eae43dbf4321dbe3437cb3b29","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-09-01 10:32:45.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-01-959001-2/FC/ : UNSTABLE in 2h 38m 14s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-01-959001-2/iSCSI/ : FAILURE in 4h 13m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"b294487136b67c10e6fe6d0e48f6426e92f42895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2025-09-01 14:00:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8bdaea471542fd9f5c9628c7ef68a71ee6b92422","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 14:22:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b8862f51d4b4229ae4e6af4fb9b939e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f5f6668886764bf29571039334458958 : SUCCESS in 20m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/30c102208bcc4949a3ae783903aa8d58 : SUCCESS in 11m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/afbad45c6b5547e5b07d14a01ac95d54 : SUCCESS in 18m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fb5db5cad3282397905032f49816a69d6a84910a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-01 14:56:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/37c967bfede0473b80fb3ea9629e8263 : SUCCESS in 49m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d64b5114cc0f49e9860b4c5c136eb7e3 : FAILURE in 54m 24s","accounts_in_message":[],"_revision_number":3},{"id":"cf0e2704a893943ecf9bfa1a28e9aabe8c0767f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-01 15:04:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1773/.","accounts_in_message":[],"_revision_number":3},{"id":"5e6b94bb58f524a37ad767b4e6454047b420eab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 15:06:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2896398b981f49b381086db53c85ee26\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9c2b2d287214c32a819d13e9e0f7f16 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95d2e93c6fd747e99c72cccd05db9609 : SUCCESS in 11m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d613058f8c874050964939414549edca : SUCCESS in 11m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bda57052c4f649e0abd8101758ba91ff : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7a0f7272fc6454eb6cffe8025d36cda : SUCCESS in 13m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/607dd1fde1ae4788bd18109edd88d2a9 : SUCCESS in 59m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/596a24d19dce4c8aa3513b6d16f74efc : TIMED_OUT in 41m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e0d491bda99445c9b311a9a12f9bf1eb : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":3},{"id":"97d9ed37658621ac06510a15de40296d3c1ae048","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 15:28:26.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/388/","accounts_in_message":[],"_revision_number":3},{"id":"37de6100113e4d1565db4342529747ddff3a12dc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-01 15:28:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959001 : FAILURE in 1h 26m 53s","accounts_in_message":[],"_revision_number":3},{"id":"99538cb1f65734e188b6dbf62e37e8e96eb8da73","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-01 15:51:31.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"d692f45b06896e5745af1b454bdc390a4b43752c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-01 16:24:10.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"f8242bb0c7b515d8940ca2c4e750be52c270bf7d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-09-01 18:54:05.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-01-959001-3/FC/ : UNSTABLE in 3h 18m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-01-959001-3/iSCSI/ : FAILURE in 4h 52m 35s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"a42954c705fc1fd38b0d2a64dfaae47a8dd0fffb","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2025-09-02 07:12:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"49488d940a1f998a050694b3bb7c3333583f9b8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 07:50:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9696ec0e435d49efbd38dac718b45bdf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/321abbee405f45cd9c9e8d4fe51b1548 : SUCCESS in 21m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f95d35a8425a4e39b8fe17cb0149b5f6 : SUCCESS in 26m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce70123c09934276b05be54a91a2cee8 : SUCCESS in 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"50e4ffdc02cf526fb605d8647bd1998cf672fa9e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-02 08:12:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5cafb871b4594dd882859744d7e316e2 : SUCCESS in 46m 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c72618469edf45a191dae6d55d9a3c03 : FAILURE in 57m 52s","accounts_in_message":[],"_revision_number":3},{"id":"b7f06bb7a32575a88d2d985c1df0888f588f31ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 08:24:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7abb89b0cd4c441fbe3ee75f7b0a49d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5559eed526994c6687ca50608a55a2c4 : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34b9cc993c7a44a4934f72b7c28c9825 : SUCCESS in 10m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/beca3e55012541b28e3ae7c7bf94cdb4 : SUCCESS in 7m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c994e16295e64ec391f8d65897830851 : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a91afaf5f8ab4b98b34802a169f29f97 : SUCCESS in 6m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9bc08302861f402280e43b621f1e54b3 : SUCCESS in 1h 06m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b283294970f84b978cb2ad92eccddaaf : SUCCESS in 16m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b726cf818a5d4de4a2f7d73a5bea2fd5 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":3},{"id":"12786e2dbdd5f98d0f434b5f4896a77bd4a6fc61","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-09-02 12:35:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4307c418437b40f37eb0eabbdc30afad7b24d24f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2025-09-08 10:45:12.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d584b24abd43bb3cf1366f4a158ebc6f1393f2a7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-08 10:46:39.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"55db8c769e89e11b214641f56904e0296b9ff400","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2025-09-08 11:11:11.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"9352e36163fca206e911ab1b5f5d3eb094123e51","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-08 11:12:38.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"4baebac4fc884bc1e3242a7c2c25058e267c9c30","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 12:38:29.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/495/","accounts_in_message":[],"_revision_number":4},{"id":"14048fe9f2187afeddb7b06d5a97e07644092b1d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 12:58:03.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"0963b9c846ba1d5b5a8f86ce681646edbc9b7b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 13:25:16.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b74bad5500a9486291d43b4723104ac2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4e96a422b9141adb79d892385801038 : FAILURE in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfec91c2d0ac40d69dd3116d279ff5dd : FAILURE in 11m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48eb2b8054cd448eb59b2319c2011a1c : FAILURE in 10m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a25fe22df0d0443db6a9692d7d15c8ea : FAILURE in 19m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f70fa82d8c4f59864d9459abfb3739 : SUCCESS in 6m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/956c4991ce1a43ac97d2150e87e699db : SUCCESS in 44m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b92b35f99a694319bb3f9497e91e4131 : SUCCESS in 1h 41m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d735413214ca4dd993319cfb87859898 : TIMED_OUT in 2h 04m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07614b091b2f4facb36b5acd8a505a2a : SUCCESS in 59m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eb6982fca8ce4a8698bce8596992f3b4 : FAILURE in 19m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5bb2948307d247db82d336b9ae96d746 : SUCCESS in 5m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e9c58634544548de99b73760f4c03b37 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c18c72708674254840494323e85342f : SUCCESS in 5m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e6bdcd73b7c1434f8b5e6258316322ac : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/10cbf4b7c0c9413c8acb114389491bcb : FAILURE in 35m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e35f22b72a7c4bebaecf95024c29dca1 : SUCCESS in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/72476f55dc064c1593a0c7993dc67d65 : FAILURE in 1h 02m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/22ef9d2ab5764da398ff5d1fdea772d1 : FAILURE in 1h 27m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e489f178c44f49e880d6290e60ff666d : SUCCESS in 1h 53m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/95c466b9dd154922beae7efac6114449 : FAILURE in 10m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9468a0deae6c4d7cac2b94ee39d7b178 : SUCCESS in 18m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/938745cee88a4168afa7084b8523a68f : SUCCESS in 53m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/973e7cfe77804d6fad519eda68acf6f9 : SUCCESS in 1h 10m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e9f4aff1e7064d538b0102120171ecd7 : SUCCESS in 37m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9c4f68ed7f954b1f8a029729f9d707aa : SUCCESS in 1h 14m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e8e4ad1b11cf444891731cab60c40902 : FAILURE in 10m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f840bfb76c1410bae1b53921e4dbe4c : SUCCESS in 1h 15m 57s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e10679702a8c4cf7b8c2cc9b0d688634 : SUCCESS in 1h 27m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/776c74b536954915b7bfeee95777a83c : SUCCESS in 1h 00m 29s","accounts_in_message":[],"_revision_number":5},{"id":"3e7cb1627af958a44678b5aac2e664513b6922b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 13:34:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54e1aa4706b3421a98f933312909de9f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b79267e9d7ad43dfb1cb44e94a6799aa : FAILURE in 11m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81cfa69deec5450c9c22b3562148f998 : FAILURE in 20m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f9a00dc6972b4e10b2fadf607ceb39a6 : FAILURE in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0b21555ecfdc52b08bb6bde45ff34fcd8f7bdeb5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 13:58:38.000000000","message":"Patch Set 5:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/499/","accounts_in_message":[],"_revision_number":5},{"id":"313d0905c04cd62c0151b5072e1baaa739fc254e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 14:18:20.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"7c0404dfda31d23afb0ad455445e2cc66a236bb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2025-09-08 14:40:38.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"7ab50a6655dd9c55601fdaa478f74e973f1eacbb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 14:40:48.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/500/","accounts_in_message":[],"_revision_number":6},{"id":"bff27a29e9ff73722a92937176a1841249082891","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-08 14:42:06.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"edfabfe525b583b72abff7f07396d2140c981745","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2025-09-08 14:45:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"667b7c495191c42d4a8cb13c21ef83ddce876280","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-08 15:01:11.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"5c8cbac2cab128aa55560d884d7462a8e9ec7672","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-08 16:47:46.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1717/.","accounts_in_message":[],"_revision_number":6},{"id":"c9d7d78faed88c800ebc5d92674dcecccb8528b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 16:59:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/edb9fe4f09b3456d93f1e2145b601c79\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c883e7e752ad456083f642d22a841a54 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b30bba10cb5441a1b0d7e31a5449e12b : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b519e57bd3246cfb28a06f6251bc18f : SUCCESS in 9m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c372c44e4b6f4fe99c254ccefb745ef4 : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99bac089a894531a0912f03356e4196 : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/a0f6b9e0ab204197bb8e3bbe5365cc49 : SUCCESS in 32m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a885e5e0c27e48b496829466cfcb26f8 : SUCCESS in 30m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/104a4ecc5d6240da8c3e4ef42a7c3f74 : SUCCESS in 1h 31m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2f87b8e66b64097bdb8972ac094c0f0 : SUCCESS in 48m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5c5f6e5f6e7442d3b48eef26d7e9a87e : SUCCESS in 19m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b9373f3077af4bf9b09a5f961ed4aac6 : SUCCESS in 4m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/10ebeb63f2924722a4b68384ac61ea53 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e2cf953a766c4977b26536e780651208 : SUCCESS in 5m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/22ee8377a7824f0f85fc41c01cb0326a : SUCCESS in 7m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4ffea02b5bec4be7b3d6ac2d0767621f : FAILURE in 26m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f36059b995e64b1eb5212cac23293e40 : SUCCESS in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b7a1037bcb048c58634b98b82250112 : SUCCESS in 1h 39m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/50554dd65cdf4a58bc449366d124751c : FAILURE in 1h 41m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a6a5ddf0e49341c8840916c5059ca446 : SUCCESS in 2h 13m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7264a1de30304147800569c7c9084c38 : FAILURE in 10m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1a2bdf202eb549b684b0475f152fa749 : SUCCESS in 32m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1ddd840b962b45a69519f926b0593109 : SUCCESS in 1h 07m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f676dc5ade5943daad68ec043707c9de : SUCCESS in 1h 37m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/996000617f604ccdbf4dc6dc3f8da756 : SUCCESS in 47m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2c1e614085f746728cf06dad0c815335 : POST_FAILURE in 2h 07m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5213719e841d40409e53c2ba96fd2990 : FAILURE in 19m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/80e251c5a1164537981520cfe749897b : SUCCESS in 1h 16m 47s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc16598387d6467dab65ec8c9246a874 : SUCCESS in 1h 41m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e2291fa2daf4e0bb034b6d2ded1bf3f : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":6},{"id":"d6f4b5e7ee4cf1a0390c55846ae5e2e5d14c3955","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-08 17:23:55.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":4},{"id":"50c25cb15232677988e2ed606522fdaea0ad829f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-08 17:27:53.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ef2383acaeb34bbdb3a94b027e69e667 : SUCCESS in 44m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d56423ec69754f99bb775d16ac88cdcd : FAILURE in 48m 16s","accounts_in_message":[],"_revision_number":6},{"id":"8699d99d95db88fadff52254a6a442d5569c16b2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-08 17:56:21.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959001 : FAILURE in 1h 15m 19s","accounts_in_message":[],"_revision_number":6},{"id":"55ce3278dfb99c56f4dc328414b94a050aa1242e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-08 18:02:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/959001 : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":6},{"id":"59bc35f9950139788c9b6871f62e0aa2648ab969","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 19:21:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db41933f181b41e2a1936324a527393f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/08350e1647e34bd79741aeb0e839f9a7 : FAILURE in 28m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0304b4674ec44bc19da4e608266dab88 : SUCCESS in 13m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aaf1bede3dce4e2592d6d90ee2891afd : SUCCESS in 24m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b3e335d9d8cd9f687a3fc4cb3a7b523561111446","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-08 19:23:46.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":5},{"id":"9ee9d9debdece493c775589605195d79ce59a703","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-08 19:25:00.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1853/.","accounts_in_message":[],"_revision_number":6},{"id":"0eaf9bc2068ae17337607a44be6201bc1ef857b9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-08 21:10:51.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":6},{"id":"ba1154851be3b7e908d2842a566ff9cd7c4634ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2025-09-10 06:38:19.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5999354171659a5b2f917d6fcaa7e74ea837f16a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 06:38:30.000000000","message":"Patch Set 7:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/520/","accounts_in_message":[],"_revision_number":7},{"id":"ddb654629aee97f77000f0b4104080bb7fb9ba0f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-10 06:58:28.000000000","message":"Patch Set 7:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/7/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"f5b8a5ee2e08a4c41f2b2e5367470a1e69603746","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 07:08:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6684f817297f42ccac3fbaf92e4fd6e5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c65292225f254316bc142d69de3099d6 : SUCCESS in 17m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dd76cf2f24c64c0c9b127a22d3350827 : SUCCESS in 22m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/659a017ee33b4311b2161ee5d8b67391 : SUCCESS in 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dc0d03583d138b5e83e5cdb822e8aa55ec979846","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-10 07:36:32.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2be75237b28b4571ad717653b967abe5 : FAILURE in 52m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/60a97d028417441bb98fe82f8f5f280c : FAILURE in 55m 21s","accounts_in_message":[],"_revision_number":7},{"id":"baf5e7e62b0035795aea9e07042945355225f99a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-10 07:42:20.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1738/.","accounts_in_message":[],"_revision_number":7},{"id":"58efbf07a67e4f5ddddecdfaf378172c90f4c068","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-10 07:42:37.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/959001 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":7},{"id":"d7908132dc61d8ae65d3731dc45bf97d2370228a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-10 07:44:42.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1872/.","accounts_in_message":[],"_revision_number":7},{"id":"cab0b2b008ae8587cbedaafc8e69a8245ab8dbc0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-10 07:54:24.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959001 : FAILURE in 1h 14m 18s","accounts_in_message":[],"_revision_number":7},{"id":"f170a65e460cd02fe52f4837cc2cf1ed87d7c6a0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-10 08:25:23.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/$JOB_ID .","accounts_in_message":[],"_revision_number":7},{"id":"b391c353e432d057565eb347199e832febaca08c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 08:35:11.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f76c9ea482c4c3699214cc0820455eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9747ae7eee064a45a3e9094d785e7b66 : FAILURE in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/502856cb8cd94ce58a9a9e4c9a4f5b53 : SUCCESS in 11m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90ef5a8b84bc44548e0ea1767d0db6a9 : SUCCESS in 7m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5dc33c0e3d1a40389356b2258a9ab8c1 : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6308740e9f9545828aaa6aa0b4e1ec6d : SUCCESS in 15m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/f82dc07d61fe476db1cab17a85980b5f : SUCCESS in 1h 04m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49c893482f9040bcb9c3ddc6c42fcb41 : SUCCESS in 39m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90f9dd31c86e474db3343d59109e111f : SUCCESS in 1h 29m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c581795e95944e49b3fc1e89c6640615 : SUCCESS in 1h 04m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d6162383acbe49c6957466c9e7b3e3d5 : SUCCESS in 20m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9cd2f4265cfa4dc190b929a7bba72c32 : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd10eaeee4c64e59b657efa6b9c8b2fe : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/305ef42aa3aa4908a8ac1b46f66b4af9 : SUCCESS in 5m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/32159dc8d6214d0c81a403c9e9626746 : SUCCESS in 8m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4ecf5eb55f654a5997795d6550d9818a : FAILURE in 35m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/efda26a5cbc84453b53b8c36afb8b07c : SUCCESS in 3m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/91da85013d8f48259ac2f520815b46ed : SUCCESS in 1h 48m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c34f46fcaa8248e0b47c354c7da9af7d : FAILURE in 1h 49m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2956cd3873d048ed8b73ed0a8a10dea6 : SUCCESS in 1h 54m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7f035f59383b4c9d841aa870c8b140ea : FAILURE in 12m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ef2112f3823445fd9e7a6dadd642181d : SUCCESS in 36m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9f2f578f523140728a401938636d4b59 : SUCCESS in 50m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/21720b631db741eb8fc1938bdd53d512 : SUCCESS in 1h 20m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3fa3aba0b46f43b98840ee81bf87307c : SUCCESS in 41m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0e688f4bdb1c4bb1a23a270e09a23042 : POST_FAILURE in 1h 38m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/31c31496f83d4273af88334b425bbc1a : FAILURE in 14m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d15985b361b9429d8eac2070938869ca : SUCCESS in 1h 22m 17s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f676968ed80a46898ea231ed8205ca49 : SUCCESS in 1h 36m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0852b684bca4aafa611d46aca60f155 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":7},{"id":"9401479640e4a82bc3881d51a346fcc939479550","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2025-09-11 04:47:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"583524d065de6accc606892d10101ed9e53cfb09","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-11 04:48:47.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"6fcd706c2f559fb2998c9b116320772687b45411","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2025-09-11 05:08:43.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"317e25069bb60c81e80b6d9cdbb9ed1a7358c25e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 05:08:54.000000000","message":"Patch Set 8:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/547/","accounts_in_message":[],"_revision_number":8},{"id":"5c7385e33ec8ce82196ebaba8905ca414eb486c5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-11 05:09:00.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa .","accounts_in_message":[],"_revision_number":8},{"id":"ffccb39c74b246b61566aa8f1d9f7a2db90a33f3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-11 05:10:10.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"61129c0ab20d626912e5bdeeace9694035897e96","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 05:27:23.000000000","message":"Patch Set 8:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959001/8/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"c0208df65be53bb515d4b952a71a16d5ff3c639d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 05:35:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f354c79c8896460f8bfffa33e379e00d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a651ac0744134684a963b2cf3ffe5b49 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b069e03cc51a4d8c818cdc493bb464f2 : SUCCESS in 16m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/081d131d06f741aba7f8efc4401df833 : SUCCESS in 24m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"51493a25b00632c9331b12efa70275434eefa339","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-09-11 05:54:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d226cac36f49af6d636abc973681e7d16786df4d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-11 06:09:56.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/959001 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":8},{"id":"4b5f33a5e3bba8e2ca6b814d7fa4ad5d3bed1223","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-11 06:12:09.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/89bc3ead2c2d4fdd992818cdc44cd80b : SUCCESS in 45m 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/67251d55d8da45db89d54840be1e0bfa : FAILURE in 59m 41s","accounts_in_message":[],"_revision_number":8},{"id":"05602985bbbb2e5a10556f662d8b77fbb9b392d5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-11 06:30:10.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959001 : FAILURE in 1h 12m 50s","accounts_in_message":[],"_revision_number":8},{"id":"eee0f2dd62c28a84059c784001f8f864c228cd33","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 07:07:55.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1891/.","accounts_in_message":[],"_revision_number":8},{"id":"9a0afc68b3dcfd5a9b042ee5b4fc3bb15a5f76dd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 07:12:51.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1757/.","accounts_in_message":[],"_revision_number":8},{"id":"0b2d61a5354120413932d19b1f80c88307940ffc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-11 07:39:38.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa .","accounts_in_message":[],"_revision_number":8},{"id":"df742d525ee0bed60e031f06d6d8fffb2e74b961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 08:00:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2110b5d1ae4449159c802bd705c7c007\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9060cd006e284749bf5f05529939de2b : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/938fe52eb84f4bf7bef8e9d2113e7f7c : SUCCESS in 10m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6501eeba831e432988de60559405ab0a : SUCCESS in 8m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8de82b32ea4a42148c27ace589f9bbe9 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8621bf612b1478da008dca9d2e17ca2 : SUCCESS in 14m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/3a2dbd19419b4cb2a39d87568a9c0673 : SUCCESS in 1h 00m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11fba89388a04d91837e20c114c37640 : SUCCESS in 1h 01m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/919478e30e2b415f8e7406ff9838933c : SUCCESS in 1h 28m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb0262d4c1bd47699742898a52a14d44 : SUCCESS in 32m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/61e2e84f2d0c452eaab1ab4b926761d9 : SUCCESS in 22m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7ac205a702434a9db606593d9c3bf704 : SUCCESS in 5m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/04090edf8edf4d2083c14bfcb9358457 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a14f4af93eb34e2883a674e8b0de15d6 : SUCCESS in 6m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b86a33b2b4c448f59ad105c3c10f9534 : SUCCESS in 5m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/081485e02dab42ff9ec0e6fce95d4faa : FAILURE in 36m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ea6ea1e450840a499945ee281be1149 : SUCCESS in 3m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2b09d0e11fb341cdb97517a86c168430 : SUCCESS in 1h 49m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d1b7cf29e2954d3f89514c4568cbae88 : FAILURE in 2h 45m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/13243f96f606494aa76c79e668884f3c : SUCCESS in 2h 00m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/92f2d5c2a0144c0a95c8f1769adc508f : FAILURE in 16m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ef453f7a76ec4af0979c4e36be71e416 : SUCCESS in 35m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3eb3a5d172e6459db3d5dd640ed5061c : SUCCESS in 57m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aed9445cf176490891e8c18e38a397bb : SUCCESS in 1h 39m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/608b799549b4466b8906e7c3df7af9a1 : SUCCESS in 42m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/65ebc56e8ad14ee0a58225777419686e : SUCCESS in 1h 10m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/de14198dbd8f4947a208da1b93621f07 : FAILURE in 19m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e5ac06064bb46ccb642ea07c2025839 : SUCCESS in 1h 36m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f6a28053a8f4d7db364d15882962ab8 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":8},{"id":"10e96ebfa0ae08730453de181a8b69373963727a","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-09-11 08:01:11.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"302c213a87ebc31020d5b377a49e409f5ed63aef","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-09-11 08:01:51.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"1ea2f38612ae8b8f2bc974e17b5cf92eaf61b548","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-09-11 08:02:32.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"a0824326202437b4be16b868bce40a2e71f1158a","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-09-11 08:03:11.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"1066d5169765695093662e02c936011d7e4fead6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2026-01-08 10:49:39.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"aa61e7bde45c9a718ad22a85ffc2eb27812c0ccb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 11:09:26.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2546041dd2e14841becde49dc16c6ed3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4d272e3dbdc48d1b3ecc97508434ff6 : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a5de031d6c464ce0ad0111d0c04c287e : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2a56dd3b1ec34d3bb05dc4a3cabd2bfd : SUCCESS in 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3c4d398814265f9c1093392a71facec0eaf66ac7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-08 11:38:04.000000000","message":"Patch Set 9:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/147/ : ABORTED","accounts_in_message":[],"_revision_number":9},{"id":"668df3da8f30439ff61d5c3e3f13f37231b657d5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-08 11:45:49.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa .","accounts_in_message":[],"_revision_number":9},{"id":"3680ef1b5c8fdf4d9c71383a4e5869f2640bb277","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-08 12:33:44.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/44cb4a80ff574b448d4c73e5b43fbfa6 : FAILURE in 39m 21s","accounts_in_message":[],"_revision_number":9},{"id":"1169040f7cd53244b99647bf2fc3bade1c9ff899","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-08 12:56:30.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/748c019d09d840d09a6b402f82bdb62a : FAILURE in 1h 12m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bdbcc752e3a7401c9169c1a7c280709a : RETRY_LIMIT in 13m 28s","accounts_in_message":[],"_revision_number":9},{"id":"3e54c4af71d96b04ec2db261e4ed9858a58ad63f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 13:26:25.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/657c58648c33426484ac386ecd8d1f1b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf0aaa134a5b4278a9bbb95cf45efb6e : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca9a4662aa7d42dc9fb1bcf35feea56c : SUCCESS in 11m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/767ac576bbf64b239c105a4e7ad0fff3 : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f39a7b4935144e71909d1ff441f6ad31 : SUCCESS in 13m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/487066e1aa204730aeba6fcc5b854364 : SUCCESS in 28m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/209821fca9b74b68bd1c686134bace4e : SUCCESS in 1h 11m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f7ce63eec1e4d9788a3998537332581 : SUCCESS in 1h 59m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23180a392a75475e8f000e0432b4f145 : SUCCESS in 1h 01m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e020e243b1df4412a3790d6d26a98de6 : SUCCESS in 16m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5ecd071941c54768bca91497b5d5a738 : SUCCESS in 23m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/66a8f69f1a1e44d09d94d73739f2b6a0 : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8d61863e2c284501a2a6875b08d47488 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c291cb9205fa49ff944249c6eaf7c2e9 : SUCCESS in 3m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/05122f1f3cb8470d873ee43904a17260 : SUCCESS in 9m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5d6ac8367eb747bbb93932288228cf6e : FAILURE in 41m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/547f6d1d873c4e6a8656c9783710a3bb : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/110728fd35ab45d1ba3ddb042664f5ca : SUCCESS in 1h 39m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b2b0d99a79f7469ea3fd47e52945e7ad : FAILURE in 1h 53m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f0f8893eca8d4f67ae5acc4fe386c113 : FAILURE in 1h 24m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f4e952b447364d7abe3642ad6249d3d6 : SUCCESS in 1h 34m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/78c68fbe99de4df1a76d96b2fef0aa77 : SUCCESS in 35m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eb893ade749c4f759027ae4038cbce3e : SUCCESS in 1h 12m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fa14af744e964bedb967d350231f5181 : SUCCESS in 1h 46m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9526cf2e85f84d40b20793b2c2b18df0 : SUCCESS in 41m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3fefa21525584ce6b6316094996b3429 : SUCCESS in 1h 27m 57s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/4ffadfbc00c544d5afa3763e39ed3ba3 : FAILURE in 2h 27m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bd38c02621b442f4a6a93c6117bdc438 : SUCCESS in 1h 34m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/596ba0b331d44e1fb0590cc4d6094cf6 : SUCCESS in 28m 26s","accounts_in_message":[],"_revision_number":9},{"id":"892f6d125b6332dd75d36229a34168ecfd210fff","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-08 15:28:13.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959001 : FAILURE in 1h 21m 27s","accounts_in_message":[],"_revision_number":9},{"id":"89f1882157a2f91a247b0c045d29f44449a24662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2026-01-09 06:10:36.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"acdddf2df45806aabc054a9a1af684b2f73ef457","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 06:28:15.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4bd19328a5542f9a7da5811612cda00\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/15e5786ed8a845a592167103625012d5 : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1ce91130019e47c9a8ec0b4a5a272017 : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/beb860a629bc4bcbacc41299ddbdb95c : SUCCESS in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"86705c349459cab67c80e547af51d50807f2ccd5","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-09 07:01:08.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/4175633fbe194669bcf92dad9d9d57d8 : FAILURE in 37m 36s","accounts_in_message":[],"_revision_number":10},{"id":"84d611e9d4bdae2580e6c89ab9e9b7fb965f37dc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-09 07:06:04.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa .","accounts_in_message":[],"_revision_number":10},{"id":"6b1a079eb59f937162531a383bdd147f4f872515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 07:16:24.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb0312a2789348a4ba24d95a6f2d7ea3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa53bad3c6f74d2f8936f6ba4f46dc53 : SUCCESS in 6m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aeda532d87f74c0496ef4101df0e2c47 : SUCCESS in 7m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3cac3cf74a59445ea3782efe66a49746 : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a75275f671f480b8156b631613d0a68 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/73e44756e8534fffa55c846db0600085 : SUCCESS in 48m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fc03dcbabb804defa57189dc2a2303a7 : SUCCESS in 28m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3fd4b4fd61594701821cd9d18f0f3aab : RETRY_LIMIT in 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88ba645d61f648c9ae6537c75d78424e : RETRY_LIMIT in 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0149fab46af342ec852b7252ac5a8148 : SUCCESS in 16m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3478f4dbfa7d4fbca40f4fa7d355f98e : SUCCESS in 21m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ae9570007f784a49bdbfe75d143647b1 : SUCCESS in 3m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5ad354fa3a1842db8f7a76e3a2edc1ce : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/08f2d590236740b5b00f712aaba5280c : SUCCESS in 5m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0768f20e884e459c8f0e79882df75181 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2d07655d3b02433e9d62e41ff82c4d68 : RETRY_LIMIT in 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/39734cd184284b0b8971147bc68d2b3b : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a8bf793cdc594c3dad97d6ad51f8432b : RETRY_LIMIT in 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/04af6362951947f5ae60a5ed6c9f9bf1 : RETRY_LIMIT in 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fbaf391ca2e449a1a636d975e94344a3 : RETRY_LIMIT in 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cba304c26aa7437da564f9fd2946d073 : RETRY_LIMIT in 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e61c12f896dc419d87e01a1a05d30c7c : RETRY_LIMIT in 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2bc24724815c424d97efd0edc468e93f : SUCCESS in 39m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/955968a9730f40d995f95d839d668175 : RETRY_LIMIT in 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3e43caf356e34c12b8fbf0c191d8d3e4 : RETRY_LIMIT in 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dffabd12cd0043ebac1237e69c4af593 : RETRY_LIMIT in 44s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/28567a0b633949e18a9b73a24df08ddd : RETRY_LIMIT in 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3e3722a1fa94d0cac02caffb51bcf2e : RETRY_LIMIT in 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a007e0876ece47c3bff8c5c25169baf3 : RETRY_LIMIT in 43s","accounts_in_message":[],"_revision_number":10},{"id":"ee26caf7411cf83e6e6036713a37c0096f62b6bd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-09 07:37:06.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959001 : FAILURE in 1h 18m 45s","accounts_in_message":[],"_revision_number":10},{"id":"8433593df840d696f3b61de49383dd208fa580d9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-09 07:42:46.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f52520fb62624423a78ced523409806f : FAILURE in 1h 27m 02s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f6f30c6104e6471c83858a0c225f2d42 : RETRY_LIMIT in 27m 23s","accounts_in_message":[],"_revision_number":10},{"id":"3f7b4b9d9fa9fec22dc015f531f5fc8b9d9c0e53","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-09 08:33:42.000000000","message":"Patch Set 10:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_160/index.html : SUCCESS in 68m","accounts_in_message":[],"_revision_number":10},{"id":"6834e5ea4256f7e66a7687b3a98271324315c908","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2026-02-17 08:50:55.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"27a7a6bf8b7a818d733f044706d08a776dbecdc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 09:19:32.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2345f90dd5e64bd8bce87a355458f7b1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1fd6765d4b31411892e5e6c569dd12f6 : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cdb43e10076049f49686c41315a94966 : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/52cc7c1d86c546a3b56d405e5b6ca4b3 : RETRY_LIMIT in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3588e73201ef7b8bd80fef85b2cdbf75dd4aa92d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 11:13:28.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52b7af6ceb0b46719ec3deb23bac9425\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ed333782da43dcbdfb22ae1fcfd18d : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62173c10249c4ae5be9b2830f7b743c3 : SUCCESS in 10m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b710c88227884d4c97d545ee2c97e5cc : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa13203c0147439a970dde12db0d89d7 : SUCCESS in 14m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/5c7925042f3c4eb693b15429c6b963d1 : SUCCESS in 32m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/823be732547c46c484560a82ab1c0b40 : SUCCESS in 59m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0408bc23a4c6449cbd2fb0496071f8f1 : SUCCESS in 1h 26m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8e69fb280804f1d8983162bb312e32a : SUCCESS in 1h 00m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2d42dd885e74e07844a51ae4d45e2ab : SUCCESS in 17m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c2896d93f2f54ef7915b53c3c7c2626c : TIMED_OUT in 40m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6450c2a94bbe4be9acb3b3dac143e029 : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/46b71cbb4e814fa1974564e52a8a5228 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/84051c8145cc47b49e7268d6cf217a00 : SUCCESS in 7m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f0bdafe154d748a9a242c08da875e801 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7770693392654f8eb4d7d03682d4a75b : FAILURE in 21m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9176a69d4f5c41449eb39eec32e70cd3 : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9276de2eb98247e4a587201dd3f39ed9 : FAILURE in 12m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f7884fe786734ae181b3e496e3c7e459 : FAILURE in 2h 17m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d9332f9d935840a1bb131b1f4b8556f9 : SUCCESS in 1h 49m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8bb47ab98e8b4a27b3d33b0625840d8e : SUCCESS in 1h 34m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0a7edfb5edc940889cb474fd3bd4969c : SUCCESS in 17m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0028402af7c043428ec44195a1841db5 : SUCCESS in 1h 00m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/90324310564f4b39a72eccbfd04e16eb : SUCCESS in 1h 48m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/43f131d7ec2045fa9e2520b8f396842e : SUCCESS in 41m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d7dd4166088e4be99ebf4e50d290796b : SUCCESS in 1h 45m 27s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/182a0e29e6ca4faab7e01b67f713d25a : SUCCESS in 2h 09m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd21810d91c64196a670a5ef83ca4db0 : SUCCESS in 1h 02m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98961d6408374381a27e0148d3daef71 : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":10},{"id":"1d60da5d11e70546a74f8263d893bb5e01c2adfb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-17 18:47:46.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dddabb272c0f4e0a8a5688bc04049fb7 : FAILURE in 14m 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3a5505a1fd094b51b891e05af06e0d66 : FAILURE in 15m 27s","accounts_in_message":[],"_revision_number":10},{"id":"fc83c1e6434df83a099d08e21b1742aa6fff8778","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-02-19 06:54:54.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fe63f038fda3b07c3c50d8d50c9aaadddeecb11e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-19 07:01:28.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/20b4d51147684e34a8e4607e0d90f40c : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/08d58a98089644ccb87e2313212847ab : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":10},{"id":"12baeef6d4cbf147ab2a66e9bb12c197114af346","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 07:12:04.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b27d637515664dcdb9dcfd6d5d1c31f0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3154abc3f0d745ef845f893c13420517 : POST_FAILURE in 11m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9881c7bc03a74c4c9c8ba81b837437ba : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0dabdd7a790e43c19419d1b740a1aa58 : RETRY_LIMIT in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fc7dc9f21869cf3f1895d888c57a1e3d941d39ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 09:29:46.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fc876a2255c485eb113d884945b21c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85326c4f8b8440a081719c4140dfa174 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3109f75f17244effa614ab492b7c9ad0 : POST_FAILURE in 7m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff9765f974344e6484550d6b5aa00032 : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee09753cd9249b7bcca331012b42eec : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/92dd504938cc4a5281dbe9215326d852 : POST_FAILURE in 34m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/13a59e9cd22f4cc2ba5f3c0c226adecb : POST_FAILURE in 1h 00m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f20129aed95462a8d07b5bc0ffc1c68 : SUCCESS in 1h 23m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce6cce5088754bc58a80a33c3fea7515 : SUCCESS in 35m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9e35cb6763845beb414c3cffafd17bc : SUCCESS in 15m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cea3e386c7b5434bbbbbec10951b15c0 : SUCCESS in 19m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/38d8292a4aca4ac7a571a6300fc24774 : SUCCESS in 6m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/32b545c099c44850a801d163d97164d1 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/829e1da483424e7882927a3852406858 : POST_FAILURE in 5m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8651e41a6e4a45d1b721feba0df9eb8d : SUCCESS in 5m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fff5ca6394de44aa8bf5780f765cdc87 : FAILURE in 41m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f620945d017495eb90da109c51ad447 : FAILURE in 3m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/15c0648ed96d4b0793a2543d8842658f : SUCCESS in 1h 49m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b49a155de6e545ca953d0238983710e7 : FAILURE in 2h 28m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/faf3d7b8e26d442e9cb149b9fdae3487 : SUCCESS in 1h 51m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1f2d1e0e8d4f45dda61bb601e3e6b7bf : SUCCESS in 1h 59m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b47e4ed7e74246a6a0ad2f8e93120fde : SUCCESS in 39m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fa3095c311d84e93a4ea41199b305a60 : POST_FAILURE in 39m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/888e4383d7cc4304a11f509222437406 : FAILURE in 1h 34m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fc9cb91b68784d898d7e32ba60b3d34a : SUCCESS in 47m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1b3124dd1a864acbb8a7c6f56c0f55d3 : SUCCESS in 2h 12m 00s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/30fc2f95ba3d485b87afeacd2da693b5 : SUCCESS in 2h 17m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/004025baefad46a99b96892e3f42247b : SUCCESS in 1h 58m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0810a4481a984ac3a56dc2cd5bfef01d : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":10},{"id":"3e649d0d54c85066446b88de35300d9a23da813b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2026-03-02 09:07:20.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"7215cb7603f3258be242c69467bb93687fa34e7e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-02 09:07:43.000000000","message":"Patch Set 11:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3116/.","accounts_in_message":[],"_revision_number":11},{"id":"7f03ba67385b33cc311783e0c2295b9bee4aca70","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-02 09:12:38.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dc4a86eb9c4248d6954ce3c5cc9eaada : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5382827402b5432bae28f8806b04c280 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":11},{"id":"e475f38d97b1317ef98aa8855f95575f1dcd38d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 09:26:07.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a345edb1600749db836124900c973893\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0cd24319ba94428093bcb3021aa7339b : FAILURE in 17m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7cf08173aef6427690a01966de61caef : RETRY_LIMIT in 4m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/04e2e403e5d5442c97aa193348e83857 : FAILURE in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d846a48ed7068c4436c4699361ea96891892902d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-02 09:46:13.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa .","accounts_in_message":[],"_revision_number":11},{"id":"eb15c094a2911c29d1fb210c4ca63c4096618c39","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-02 09:58:55.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/62df762c503b4597a6f22ce90fb74211 : FAILURE in 41m 41s","accounts_in_message":[],"_revision_number":11},{"id":"cebcc719672742a232c75b0cb56e47bd9fcf21df","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-02 10:12:44.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver build/bf8d6ca034e74d969bda37422928a390 : FAILURE in 33m 07s","accounts_in_message":[],"_revision_number":11},{"id":"abafd5a93a0a9f48bf6869c46223e47d55f4d0e8","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-02 10:28:23.000000000","message":"Patch Set 11:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/959001 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"387f0aee3d132c9fda15e3ab55a0b8ab22ac518b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-02 10:35:59.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959001 : FAILURE in 38m 29s","accounts_in_message":[],"_revision_number":11},{"id":"e77f7380b07910e27d084e02a24e04bcd72600fd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-02 10:38:24.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/02c25e9ad2094c97bde73ef6a47f71fa : FAILURE in 37m 40s","accounts_in_message":[],"_revision_number":11},{"id":"62c13750d22d650873f6cb401e2fca34d7845e7b","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-02 10:39:16.000000000","message":"Patch Set 11:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959001 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"a3025a146b9c885deefe2108ad770bd900beeef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 11:31:28.000000000","message":"Patch Set 11: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf3ec0e5dc2642138409e955d48023ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c167c3fe3d7a4b72adf99162ced13794 : FAILURE in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54deafc984bb4800a512a149670d6e88 : FAILURE in 10m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8933ca7fe7e84c9ab56ccd331f1feaea : FAILURE in 10m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f9b164b296384c248d4519cd80b55963 : FAILURE in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f06f5ad4a6144534a4bd9004c1dbcccd : SUCCESS in 10m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/20e205ef8f8a42c3953aeef358f1780d : SUCCESS in 1h 05m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ebfa93050e0e4258bd17cbd6b4dee29f : SUCCESS in 1h 01m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7fb5229ade7d4864910bebccbc5eca84 : SUCCESS in 1h 41m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dde1286027da466ab3ab642886b25c1c : SUCCESS in 33m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a658d7d0161f4e79a8c2d45f5803b0ee : SUCCESS in 17m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2cedb65f271a44d6b68df3ab122324cf : FAILURE in 25m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/21a8c1e632394f89a5b66a0796849cd4 : SUCCESS in 6m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7ddc31a1a8f34c79be82352d0ca74916 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/423bad296cb940378ee6d595b4b150fc : SUCCESS in 6m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/44b6a2ba73d546d3ba5cd5a5abe411f1 : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/396d1250ba3f4b38b648acb45f2236ba : FAILURE in 42m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4377ff58cb2f4de29823ba2de0714a7c : FAILURE in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e0d1fda8989040b489e80191ebfcbf54 : SUCCESS in 2h 16m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3c9646f89847425dbae54d7dc01cbba7 : FAILURE in 2h 17m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/66759b80ef724c96b228ededa43c20a1 : SUCCESS in 1h 57m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/de9b4decd38b46039225e7fed47ac0a1 : FAILURE in 18m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/288cb2b598144b8ca032b931c5cb4d03 : SUCCESS in 38m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/77d1337f7225420c86efbe50ef7c8477 : SUCCESS in 1h 08m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c5b0640088694b8e881440c9be6651ba : SUCCESS in 1h 57m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5312d7db810645af9522f8368141d514 : SUCCESS in 42m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/deb3c1b7ac264cb0a44c76e4d8d1fc11 : SUCCESS in 2h 07m 11s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/7ea784c0aa874552ba1e8dda145ab52b : FAILURE in 14m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f0d5a2d42184ea091910389a771631d : SUCCESS in 1h 53m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c44eb6d868564976a324625dd267c405 : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":11},{"id":"9d26c8cb8bf90ac8a8751b308ef12f01ead496b9","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-02 14:38:13.000000000","message":"Patch Set 11:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_912/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":11},{"id":"18c2d07bccd7f814869777c5dd5206a903cb1946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-05-05 11:00:49.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"5c71a2bdab06918fbabfac10924d2d3c09ac8488","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 11:18:46.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18708f65b6424d159f7ac4530e189e67\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/037716b89a0d4e0eacfe8f787d5a4d4e : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c211562f74ac46c090c659e846020690 : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5445327d596d40bbaa95cfb4cedea212 : SUCCESS in 16m 06s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f9df50ee26b74f6b6ae05004e5eda1e75215947f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-05 12:01:07.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa .","accounts_in_message":[],"_revision_number":12},{"id":"9186f3d6e2d8a0d0583ac702bbcddaea2ecf5453","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-05 12:06:52.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver build/d4e5ba7ac221479a90cb77958d07f6fd : FAILURE in 30m 46s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/959001/etc/tempest_log.log","accounts_in_message":[],"_revision_number":12},{"id":"ca9685be8c4ec74f30adb6867ac5dad9c61a7a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 12:10:10.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb5a1f7d4d3f48dfb27d274dea8e7db5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/989d05ba60f84d5da427fe8a85bbb767 : SUCCESS in 6m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ca10e9156a74506a4324d5c2c93bce7 : SUCCESS in 12m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a4601a446204bb18e7b5663f4ab066b : SUCCESS in 11m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b0c4970a931c43828aa53d9353689054 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19f3b39fa9b24d5eb77db91f85364943 : SUCCESS in 11m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/523230c4bdaa470e8146b823bf70d717 : SUCCESS in 1h 03m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91f9fdeb7b5d464cb9641c2968c9aa0c : SUCCESS in 17m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/41599cde4806438c9df5ad6ba3974a8e : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2ecb1775ff0649a28a00feffa4552018 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":12},{"id":"fa54be0fb20b8877ac56af9b23c39c46217d6c53","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-05 12:13:12.000000000","message":"Patch Set 12:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/959001 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"375cd82241f273e32c2ecae7e76deb984af088dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-05 12:19:43.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_05_2026_12_13_39\nLeave a comment with","accounts_in_message":[],"_revision_number":12},{"id":"6caec8ad82bd33810a2469108ee8aeb0265c5b59","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-05 12:21:29.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-959001-12-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-959001-12-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"5e08b2dd620dbcf74f647d250c57d0ca5b589e5b","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-05 12:27:19.000000000","message":"Patch Set 12:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959001 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"bb5f60717ea452f2428dbac6bb21d39317f5feba","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-05 13:10:22.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3896/.","accounts_in_message":[],"_revision_number":12},{"id":"39adeb3cde3cafcca1026e3abfd862225c20afd7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-05 13:36:46.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4072/.","accounts_in_message":[],"_revision_number":12},{"id":"08bb137ae501482ee10464ef36e963606832b87d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-15 21:48:03.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/22842cf3b3df41f2acf8f2f95fcc9301 : FAILURE in 28m 38s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959001/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959001/etc/tempest_log.log","accounts_in_message":[],"_revision_number":12},{"id":"b0dd7cd0cfcec60e5d2cde46b447cd3fe74e6840","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-06-18 12:56:11.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"83cea86e0cea8abc927cf38dc20972f2ebff2991","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-18 13:56:00.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa .","accounts_in_message":[],"_revision_number":13},{"id":"b4ec41037752ebde42753b8605b6c676e1c32f84","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 13:59:59.000000000","message":"Patch Set 13:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/500 : SUCCESS in 0h 39m 52s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/959001","accounts_in_message":[],"_revision_number":13},{"id":"8bcbecb17c266b9e1c52c577fa96d3c7449397bc","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 14:03:31.000000000","message":"Patch Set 13:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/232 : SUCCESS in 0h 40m 27s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/959001","accounts_in_message":[],"_revision_number":13},{"id":"7afc09b8d2f0d0bc8468a46137fb89e7bdec4387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 14:04:39.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4a306256dc94778b0bc71653deedee8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd63fe57b5054c518b5aa4abff0ebc61 : SUCCESS in 7m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f71d616bdee64486b528856dcf58c9ee : SUCCESS in 9m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b79e0ce32d24451d9fc233ee4f8820a5 : SUCCESS in 11m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a92beb33f5e34547a90f19e6bfa262d6 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fac433e5c79b4ba5b3c0f532bcec3877 : SUCCESS in 7m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c31736f479cd45fd82016bce734faee1 : SUCCESS in 1h 02m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db3f7ffa8ece4d0480506d9eeec93cc7 : SUCCESS in 9m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b22bd5b230ba49d49512fa85d4e42426 : SUCCESS in 17m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/16c395813352489281ca8e991eae95a7 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":13},{"id":"581e402ed5c77ee19168015b7085840822852795","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 14:05:54.000000000","message":"Patch Set 13:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/146 : SUCCESS in 0h 43m 58s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/959001","accounts_in_message":[],"_revision_number":13},{"id":"42b9a15167ab2d27484b004e5b8e305228e98bb8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-18 14:15:42.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_06_2026_14_08_05\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"d678f41c12dd361b76f6ce220b8080ba19da2275","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-18 14:21:27.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- hpe-fc-driver build/2d098f9b79bf4cd580c7e52a53a97b25 : SUCCESS in 34m 57s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/959001","accounts_in_message":[],"_revision_number":13},{"id":"675202b3b6b99e325576a09ce6bc029ccf706f95","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-18 14:34:20.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/f91808ffec554895bdb8e8b5943d707b : SUCCESS in 47m 25s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/959001","accounts_in_message":[],"_revision_number":13},{"id":"452196314876717ff6bdddaa02ace095aa40b65a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-18 14:44:17.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2d6fa951d321445f8f273ca78809684c : SUCCESS in 40m 25s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959001\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959001","accounts_in_message":[],"_revision_number":13},{"id":"b01414f033ea100cf66ecdd8cb2d6674ccef111d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 15:00:29.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14b17a889e974a509ae1aef64c7c51f3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a6b4dee842a4f5785ac85343ec81051 : SUCCESS in 17m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/772c902584c84fd3974b77769e8253c6 : SUCCESS in 16m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c648e09d3dd740dfb1732587bf95f7e7 : TIMED_OUT in 1h 01m 53s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"687c15c6a113043714cf548752133a5083f3fe1e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-18 15:41:40.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest build/c28f9daf25ad44e0a1f832dcb841ae4a : FAILURE in 51m 17s\n- cinder-storpool-tempest-iscsi-multipath build/c39bd85fd8ac493aa6ea4edcd5628cf3 : SUCCESS in 57m 53s","accounts_in_message":[],"_revision_number":13},{"id":"56631686867bb3ae8926e2512918a940eccf9f6a","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-19 03:09:56.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-959001-13-fc : SUCCESS in 1h 33s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-959001-13-iscsi : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":13},{"id":"a70855170d8f464df8f4515a08412bd5fcb04b75","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:00:21.000000000","message":"Patch Set 12:\n\nBuild Failed","accounts_in_message":[],"_revision_number":12}],"current_revision_number":13,"current_revision":"f4c4609c1234ba6feb5c3bee8c3c2751cadd6a83","revisions":{"26f47ae76d92bb54dbc518889024908bfea40c9c":{"kind":"REWORK","_number":1,"created":"2025-09-01 06:10:47.000000000","uploader":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"ref":"refs/changes/01/959001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/1"}}},"commit":{"parents":[{"commit":"2753c8a5ca67deab0727226887e2ef5386f8a3e3","subject":"Merge \"[Pure Storage] Fix snapshot replication interval issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2753c8a5ca67deab0727226887e2ef5386f8a3e3"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 06:03:51.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26f47ae76d92bb54dbc518889024908bfea40c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26f47ae76d92bb54dbc518889024908bfea40c9c"}]},"branch":"refs/heads/master"},"ae4d8213058397d6088d841a76dfc0843f85e73c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-09-01 06:18:03.000000000","uploader":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"ref":"refs/changes/01/959001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/2"}}},"commit":{"parents":[{"commit":"2753c8a5ca67deab0727226887e2ef5386f8a3e3","subject":"Merge \"[Pure Storage] Fix snapshot replication interval issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2753c8a5ca67deab0727226887e2ef5386f8a3e3"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"Prasant Padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 06:18:03.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasant Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae4d8213058397d6088d841a76dfc0843f85e73c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae4d8213058397d6088d841a76dfc0843f85e73c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6a92418e103de330b704ac1d03f761216e002561":{"kind":"REWORK","_number":3,"created":"2025-09-01 14:00:28.000000000","uploader":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"ref":"refs/changes/01/959001/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/3"}}},"commit":{"parents":[{"commit":"2753c8a5ca67deab0727226887e2ef5386f8a3e3","subject":"Merge \"[Pure Storage] Fix snapshot replication interval issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2753c8a5ca67deab0727226887e2ef5386f8a3e3"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"Prasasnt Padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 13:59:48.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6a92418e103de330b704ac1d03f761216e002561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6a92418e103de330b704ac1d03f761216e002561"}]},"branch":"refs/heads/master"},"f0b7cb13fb17776ee21b6e91675e7d73a11240b0":{"kind":"REWORK","_number":4,"created":"2025-09-08 10:45:12.000000000","uploader":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"ref":"refs/changes/01/959001/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/4"}}},"commit":{"parents":[{"commit":"2753c8a5ca67deab0727226887e2ef5386f8a3e3","subject":"Merge \"[Pure Storage] Fix snapshot replication interval issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2753c8a5ca67deab0727226887e2ef5386f8a3e3"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"Prasasnt Padhi","email":"prasant.padhi@dell.com","date":"2025-09-08 10:44:29.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0b7cb13fb17776ee21b6e91675e7d73a11240b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0b7cb13fb17776ee21b6e91675e7d73a11240b0"}]},"branch":"refs/heads/master"},"0ce5cc9a7280dbb1e9449a38b1864acceb3f2745":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-09-08 11:11:11.000000000","uploader":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"ref":"refs/changes/01/959001/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/5"}}},"commit":{"parents":[{"commit":"35ef9616b307f970c63d2d0c0a39b8ef38817409","subject":"Merge \"Fix remaining UUIDWarning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/35ef9616b307f970c63d2d0c0a39b8ef38817409"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"Prasasnt Padhi","email":"prasant.padhi@dell.com","date":"2025-09-08 11:08:39.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ce5cc9a7280dbb1e9449a38b1864acceb3f2745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ce5cc9a7280dbb1e9449a38b1864acceb3f2745"}]},"branch":"refs/heads/master"},"88d4bb9f44896ba971bef51ea4c29783b9b793f2":{"kind":"REWORK","_number":6,"created":"2025-09-08 14:40:38.000000000","uploader":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"ref":"refs/changes/01/959001/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/6"}}},"commit":{"parents":[{"commit":"35ef9616b307f970c63d2d0c0a39b8ef38817409","subject":"Merge \"Fix remaining UUIDWarning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/35ef9616b307f970c63d2d0c0a39b8ef38817409"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"Prasasnt Padhi","email":"prasant.padhi@dell.com","date":"2025-09-08 14:28:48.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88d4bb9f44896ba971bef51ea4c29783b9b793f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88d4bb9f44896ba971bef51ea4c29783b9b793f2"}]},"branch":"refs/heads/master"},"33769fd014dc7580647126e18d1072d0637f8805":{"kind":"REWORK","_number":7,"created":"2025-09-10 06:38:19.000000000","uploader":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"ref":"refs/changes/01/959001/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/7"}}},"commit":{"parents":[{"commit":"dd04417ad559eb3db5bff732a5db50df3a8ca729","subject":"Merge \"Register glance user in keystoneauth plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd04417ad559eb3db5bff732a5db50df3a8ca729"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"Prasasnt Padhi","email":"prasant.padhi@dell.com","date":"2025-09-10 06:37:42.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33769fd014dc7580647126e18d1072d0637f8805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33769fd014dc7580647126e18d1072d0637f8805"}]},"branch":"refs/heads/master"},"db61002bbed407d00652931852d99d1b4d9bc2c4":{"kind":"REWORK","_number":8,"created":"2025-09-11 05:08:43.000000000","uploader":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"ref":"refs/changes/01/959001/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/8"}}},"commit":{"parents":[{"commit":"fe5861dc19db2b67845b83f50d8fd56a9b978ddf","subject":"Merge \"[Pure Storage] Consistent hostnames across replicated backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe5861dc19db2b67845b83f50d8fd56a9b978ddf"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"Prasasnt Padhi","email":"prasant.padhi@dell.com","date":"2025-09-11 05:08:14.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db61002bbed407d00652931852d99d1b4d9bc2c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db61002bbed407d00652931852d99d1b4d9bc2c4"}]},"branch":"refs/heads/master"},"bcd9e97d6e71d81abe98b9e1ff940a486c78e2c9":{"kind":"REWORK","_number":9,"created":"2026-01-08 10:49:39.000000000","uploader":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"ref":"refs/changes/01/959001/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/9"}}},"commit":{"parents":[{"commit":"fe5861dc19db2b67845b83f50d8fd56a9b978ddf","subject":"Merge \"[Pure Storage] Consistent hostnames across replicated backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe5861dc19db2b67845b83f50d8fd56a9b978ddf"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"Prasasnt Padhi","email":"prasant.padhi@dell.com","date":"2026-01-08 10:49:12.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bcd9e97d6e71d81abe98b9e1ff940a486c78e2c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bcd9e97d6e71d81abe98b9e1ff940a486c78e2c9"}]},"branch":"refs/heads/master"},"350a1872ff31023c758f9a14cdda08810bd32caa":{"kind":"REWORK","_number":10,"created":"2026-01-09 06:10:36.000000000","uploader":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"ref":"refs/changes/01/959001/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/10"}}},"commit":{"parents":[{"commit":"fe5861dc19db2b67845b83f50d8fd56a9b978ddf","subject":"Merge \"[Pure Storage] Consistent hostnames across replicated backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe5861dc19db2b67845b83f50d8fd56a9b978ddf"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"Prasasnt Padhi","email":"prasant.padhi@dell.com","date":"2026-01-09 06:10:08.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/350a1872ff31023c758f9a14cdda08810bd32caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/350a1872ff31023c758f9a14cdda08810bd32caa"}]},"branch":"refs/heads/master"},"b54881108e88854772a10d8875ea79ae198a0aa0":{"kind":"REWORK","_number":11,"created":"2026-03-02 09:07:20.000000000","uploader":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"ref":"refs/changes/01/959001/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/11"}}},"commit":{"parents":[{"commit":"741becdcda17c49d96c7e957103ab5e74516f2e0","subject":"Hitachi: Use override_config in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/741becdcda17c49d96c7e957103ab5e74516f2e0"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"Prasant Padhi","email":"prasant.padhi@dell.com","date":"2026-03-02 09:06:57.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b54881108e88854772a10d8875ea79ae198a0aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b54881108e88854772a10d8875ea79ae198a0aa0"}]},"branch":"refs/heads/master"},"4552ebac3274ba6ff37e389e11453a823746bd79":{"kind":"REWORK","_number":12,"created":"2026-05-05 11:00:49.000000000","uploader":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"ref":"refs/changes/01/959001/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/12"}}},"commit":{"parents":[{"commit":"849ddb6cef2b06084b7b116cb2ae5ce7da610e1b","subject":"Merge \"Use futurist-based thread pool for volume manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/849ddb6cef2b06084b7b116cb2ae5ce7da610e1b"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"Siddharth Kumar","email":"siddharth.kumar1@dell.com","date":"2026-05-05 09:16:04.000000000","tz":0},"subject":"Add support for Dell Powerstore driver UT coverage","message":"Add support for Dell Powerstore driver UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4552ebac3274ba6ff37e389e11453a823746bd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4552ebac3274ba6ff37e389e11453a823746bd79"}]},"branch":"refs/heads/master"},"f4c4609c1234ba6feb5c3bee8c3c2751cadd6a83":{"kind":"REWORK","_number":13,"created":"2026-06-18 12:56:11.000000000","uploader":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"ref":"refs/changes/01/959001/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/959001/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/959001/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/959001/13"}}},"commit":{"parents":[{"commit":"849ddb6cef2b06084b7b116cb2ae5ce7da610e1b","subject":"Merge \"Use futurist-based thread pool for volume manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/849ddb6cef2b06084b7b116cb2ae5ce7da610e1b"}]}],"author":{"name":"prasasnt-padhi","email":"prasant.padhi@dell.com","date":"2025-09-01 05:40:51.000000000","tz":0},"committer":{"name":"siddhvrth","email":"siddharth.kumar1@dell.com","date":"2026-06-18 12:54:56.000000000","tz":0},"subject":"Dell Powerstore: Improve UT coverage","message":"Dell Powerstore: Improve UT coverage\n\nThis patch updates the Cinder Dell Powerstore driver Unit Test coverage to achive 85%.\n\nImplements: blueprint ut-improvement-powerstore-driver\nChange-Id: Ib6ed01f185fc4ef1965fa35e315e16662ee0d1aa\nSigned-off-by: Prasasnt Padhi \u003cprasant.padhi@dell.com\u003e\nCo-Authored-By: Siddharth Kumar \u003csiddharth.kumar1@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4c4609c1234ba6feb5c3bee8c3c2751cadd6a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4c4609c1234ba6feb5c3bee8c3c2751cadd6a83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
