)]}'
{"id":"openstack%2Fcinder~959105","triplet_id":"openstack%2Fcinder~master~Ic654fe7be872345785c29914a12206d6e3c49a2a","project":"openstack/cinder","branch":"master","topic":"bp/improve-powermax-UT","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-01-22 09:16:27.000000000","reason":"\u003cGERRIT_ACCOUNT_37784\u003e replied on the change","reason_account":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"}},"36132":{"account":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"last_update":"2026-04-20 10:03:09.000000000","reason":"Someone else replied on the change"},"37784":{"account":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"last_update":"2026-04-20 10:03:09.000000000","reason":"Someone else replied on the change"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-01-28 20:16:15.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"38286":{"account":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"last_update":"2026-07-13 12:51:58.000000000","reason":"Someone else replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-01-22 09:16:27.000000000","reason":"\u003cGERRIT_ACCOUNT_37784\u003e replied on the change","reason_account":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"}}},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-01-22 09:16:27.000000000","reason":"\u003cGERRIT_ACCOUNT_37784\u003e replied on the change","reason_account":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-01-22 09:16:27.000000000","reason":"\u003cGERRIT_ACCOUNT_37784\u003e replied on the change","reason_account":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"}},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-07-15 10:03:21.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"38702":{"account":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"last_update":"2026-04-20 16:17:56.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic654fe7be872345785c29914a12206d6e3c49a2a","subject":"Dell PowerMax: Improve UT coverage","status":"NEW","created":"2025-09-02 05:27:38.000000000","updated":"2026-09-10 15:48:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2a4dcb6b41dddcb76e86ac1cd7e2b9ad3599a9f7","_number":959105,"virtual_id_number":959105,"owner":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-13 13:30:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-07-15 10:03:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-02 05:27:50.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-09-02 05:29:06.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-02 05:51:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-02 06:25:16.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-02 06:27:56.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-09-02 07:02:30.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-09-02 07:14:01.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-02 07:35:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-02 08:13:59.000000000","updated_by":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"reviewer":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"state":"CC"},{"updated":"2025-09-02 17:03:17.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-09-08 00:33:28.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-22 08:48:42.000000000","updated_by":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"reviewer":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"state":"CC"},{"updated":"2026-01-22 08:48:42.000000000","updated_by":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-01-22 08:48:42.000000000","updated_by":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-01-22 09:16:27.000000000","updated_by":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-01-22 09:16:27.000000000","updated_by":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-01-26 15:23:16.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-01-28 20:16:25.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-01-28 21:32:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REMOVED"},{"updated":"2026-01-28 22:15:48.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-29 19:29:32.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-01-30 14:12:15.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-02-05 21:44:52.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2026-04-07 05:04:40.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"state":"CC"},{"updated":"2026-04-08 16:47:03.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-17 21:46:42.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-04-17 21:50:16.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-17 21:52:04.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-04-20 16:17:35.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-05-29 17:37:02.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-29 19:55:51.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-06-18 20:21:26.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-13 12:52:25.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-07-13 12:53:19.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-09-09 12:45:26.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2026-09-10 15:48:52.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"}],"messages":[{"id":"a464a3177404e2f67babea53d7d9adee27bc150f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-09-02 05:27:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87a5241882eeaef17d3695422ef9d1df3872ef83","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-02 05:27:50.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/390/","accounts_in_message":[],"_revision_number":1},{"id":"0177cf647156b6f0e1a412b5a9aaf45656fb446a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-02 05:29:06.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"3290edbb752f27a88dbf4751f905a60547b3b0b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 05:51:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c25bc2c5e15431db12fb280d0bdcc18\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1f8426f6417849258b4f179e727f86b4 : SUCCESS in 20m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/39e435493a5940888cf95e56b824b1c4 : SUCCESS in 20m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ecb416ff084c4835b6c9921ad94a505c : SUCCESS in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21a701ac995c72deaedb1201f7559c9174f208a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-02 06:20:57.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"13ad635ccc9493809341b1339d1d39104a249f9d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-02 06:25:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/66afa239619c4047a2dd796ff0a875e4 : SUCCESS in 47m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a87659eb3cf34b68880c03fdaeb07801 : FAILURE in 55m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d21b736390cc60445bb4678cd643d58b4c74caec","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-02 06:27:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1777/.","accounts_in_message":[],"_revision_number":1},{"id":"2ca6c9b0b23b87450923b5688368f2baec05ddc1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-02 07:02:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959105 : FAILURE in 1h 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1145d7e6969c7e89af2678dcb706566aa6e8c9a9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-02 07:14:01.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e533bdad2d995dd19f241eb1db41a910c75ebf1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 07:35:45.000000000","message":"Patch Set 1: Verified-1\n\n(117 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5019ca307e4248c283d3d039ddc8c71c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75763d6100f24772a04f17c542a06e5f : FAILURE in 6m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f93c12d397c044628b4169e855aff1f2 : SUCCESS in 11m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/afc82b0a1d834b8faa4bc4215273877a : FAILURE in 10m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcce575ac5114cf2a950b30bf195083f : SUCCESS in 15m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aba529f4867b4df6a09918650dac011b : SUCCESS in 11m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/e26338d2e236451c9a28664fc9cab0d9 : SUCCESS in 29m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ff0685991fe646b4b9fe5991e4c53e0a : SUCCESS in 1h 05m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e67cd7306cd4c95852c656b5a59811b : SUCCESS in 1h 30m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3418cc2c01740c68c00781d93b4ba9e : SUCCESS in 44m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/591d29e99eb740ed98e407d931a2da6d : SUCCESS in 20m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0df5ad7a674e4be6ace149282816f889 : SUCCESS in 3m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f669698970e4a84bce21509e8ffe294 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc3af7e48ab242f8b8b887cfc78a429b : SUCCESS in 5m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fed96cf8928548d9b2a6d4d05d6221c3 : SUCCESS in 7m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a872cfde8f2a4919a8e5e183748eb354 : FAILURE in 27m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e17fceba92445ba93b22a6d9ac3951d : SUCCESS in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/03d3bfbf87f846c2a21578420a7b1489 : SUCCESS in 1h 50m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ff82dcf0502342b3a8e5472bdfa25778 : FAILURE in 1h 41m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/14c8663bd13644939212535382c83ed9 : FAILURE in 2h 06m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c22af58570c14e53bd659477271de862 : FAILURE in 11m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cfa711d552964fd79e399910df0be8c1 : SUCCESS in 33m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c32d33b9b0ad41da902c5bf63a2737cc : SUCCESS in 1h 13m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/106c56931ea3433d99b0881bfa4f6bb4 : SUCCESS in 1h 08m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/812ce9a170894815afd38e4dd89ae127 : SUCCESS in 44m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/82ce3468c3664aab9273e2c7504a671e : SUCCESS in 1h 35m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/02598ee10ee74f7eba1950ec4962bc3d : FAILURE in 14m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bc330ed2446a4634bc67f45a344c10fb : SUCCESS in 1h 35m 09s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afe455a213d642bd8a1c6828b5f9cb8c : SUCCESS in 1h 30m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6a676014a6d4409beae35d70615d1ed : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":1},{"id":"850f81af4de328d081217699672c9aed349434ef","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-02 07:40:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1642/.","accounts_in_message":[],"_revision_number":1},{"id":"0a521dbc1c0f29456ca12359cae75d67691180f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2025-09-02 07:55:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"506d838dfccb21e2a246f32129f74944c85a800f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-02 07:56:05.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/391/","accounts_in_message":[],"_revision_number":2},{"id":"e3679f4012c3302aeef9d8e88a8f550321930d5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 08:28:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0d476a691ae45f0a04c810875edf217\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5904d1c1cc3d437db4a6fcf7576492d4 : FAILURE in 23m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6762215dd5624121bdb417c22ad5b276 : FAILURE in 24m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/781d5fb249f44232b31fa0a90cfc7885 : SUCCESS in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d38149724bca494c1574d60012cdcd94b1d6919c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 08:46:08.000000000","message":"Patch Set 2: Verified-1\n\n(129 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db32f8e11b4e4e4990cf2dd9ee0888c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cffed9d69f894252b32a28b8d1533e32 : FAILURE in 8m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a5b0806880b4e27af500b593125f53a : FAILURE in 7m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bb37f2b1abd49ceae9dc4efa9a111d2 : SUCCESS in 12m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3ab2fd85979943e3aa97f2bd61ba9ae6 : FAILURE in 14m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2993586e733d40729fe1f9adc5a72698 : SUCCESS in 7m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/880da6677a654a8b851fc195cae76703 : SUCCESS in 48m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/13037923842a4fbb8ce5450351bb3fc8 : FAILURE in 17m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cf4dcc959724403e8d8e0d4c01163ef3 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"e2f8510b7a4424d09a15d12ab22e08b70a80dcf6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-02 08:49:44.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"ed41f664120d3fe563775ff6fb95b5a585218860","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-02 08:55:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1643/.","accounts_in_message":[],"_revision_number":2},{"id":"4e02a36c9ebd35549c6858cb20ffd619f1d41ac5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-02 08:58:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1778/.","accounts_in_message":[],"_revision_number":2},{"id":"50ee287de8fac0b40aceb6063eb6f4ac361745e0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-02 09:21:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959105 : FAILURE in 1h 24m 34s","accounts_in_message":[],"_revision_number":2},{"id":"6d0906365acf3d07a1a0fb730662ddba9626edd5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-02 09:23:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dc17de1e706d4c579a3f349949374c80 : SUCCESS in 46m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7589511e3d0d4b01964414059ce88401 : FAILURE in 1h 10m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2cb6f61e2fa60e44b7e3eeb66b64691b8607e750","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-02 09:45:29.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"ced3982c6fbaf311a64230938ed4cf75d7d9c587","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-09-02 17:03:17.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-959105-1/FC/ : FAILURE in 11h 34m 37s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-959105-1/iSCSI/ : FAILURE in 3h 56m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"4e8979ff92867ed3c5d36b32bc9f9161784c9a61","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-09-02 19:49:16.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-959105-2/FC/ : FAILURE in 11h 52m 6s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-959105-2/iSCSI/ : FAILURE in 3h 49m 43s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"f87b5daffb30815bf950ff59372bda034da49bf9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-09-08 00:33:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/959105/2/20250908000707/NEC-ISCSI : SUCCESS in 5h 43m 57s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/959105/2/20250908000707/NEC-FC : SUCCESS in 9h 26m 19s","accounts_in_message":[],"_revision_number":2},{"id":"55ae37b07e7c37b70b014e82a6a11975564139b7","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-01-22 08:49:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c6d088ebd9f24ea3535a8a2b653fdd0720dc257","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-01-26 15:23:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4cf9fe9c1d4b7be55e452980388ef80a64e9d616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-01-28 20:16:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5686f5f5c653fe609e6ee0cded8451ebd00f481d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-28 20:16:25.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/716/","accounts_in_message":[],"_revision_number":3},{"id":"0479d3eaa9100c8a9b77f1b933082e945b2d5e46","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-28 20:16:29.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"ea9f9e6d5ab58ac119edae19f104035921f3b18d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-28 21:17:44.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ic654fe7be872345785c29914a12206d6e3c49a2a .","accounts_in_message":[],"_revision_number":3},{"id":"1878e11417babe455a7231374062845dce1faeb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 21:19:01.000000000","message":"Patch Set 3: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b997c7172bd4262945bedbb934ced41\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cee27e137b5b4f35bbb7925f66c00192 : FAILURE in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c809759d62b4f358c76a192613e0245 : FAILURE in 8m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e3bbeb47d224eb2b0ff7d27a31d3664 : FAILURE in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4b53f4305e4fb99f1a086c9b807095 : SUCCESS in 14m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf1a8d7c766746fc8834c7d52602d2a7 : SUCCESS in 1h 02m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c876574b4e10488081ea6335cb820caa : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ba3faff5f77a4544bc0adf8101596988 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":3},{"id":"376aa1469ec39b7e8fd7b18db9dff768723bc380","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-01-28 21:32:49.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_16643\u003e.","accounts_in_message":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"_revision_number":3},{"id":"bcde1836e409bb19cb9cce5e6f80ac522c30b64d","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-28 22:15:48.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/453aa2118359432685c248c9896535f7 : FAILURE in 39m 15s","accounts_in_message":[],"_revision_number":3},{"id":"bacee122990e706e75d68232227e749d49893cdb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-29 02:56:55.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959105 : FAILURE in 39m 29s","accounts_in_message":[],"_revision_number":3},{"id":"e486bc26abb8f694f37ed62a7a83891dbc922e4f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-29 07:29:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ced378dbaa4949b28f8dbbc8931aa35a : SUCCESS in 46m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5c7be4f6d2524cfe947827736d9b5329 : RETRY_LIMIT in 15m 00s","accounts_in_message":[],"_revision_number":3},{"id":"c59f0e35858705933151213a968331bde677d080","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-01-29 09:17:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8788a9b12db1ca32ae68e662d6773403b7863962","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 09:17:48.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/722/","accounts_in_message":[],"_revision_number":4},{"id":"96fb44d371cb682a0b68a5808af41446a29ba779","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 09:17:51.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"dc197b9ed3e68f4ac6f658990d63417a9a943f91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 09:48:08.000000000","message":"Patch Set 4: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/377d056c5585450d80c9881facd460b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abf10d4d72114faaae19e7bae6c2f87a : FAILURE in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/378862a84e444640a460c961443998b4 : FAILURE in 11m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce3b96ebcd2447efb2cf330107adf84e : FAILURE in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da45ec4acd264de2ba8449aede9e74b3 : SUCCESS in 6m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/22a45c8e8da9441597c8d815f56f1f58 : SUCCESS in 28m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/68458e643f5b428a8bf3e40e2b09f5de : FAILURE in 20m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f05d191c70884a47862e6979af898fbb : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":4},{"id":"0a951c71810bc95f34f898f9c3ba965f1a57bb34","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-29 10:13:49.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic654fe7be872345785c29914a12206d6e3c49a2a .","accounts_in_message":[],"_revision_number":4},{"id":"86abb362568fb350a2d70b2664e017a0baf4fce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-01-29 10:26:35.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"94aff4d3632de2855a48804ca1dbff0f5de531f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 10:26:43.000000000","message":"Patch Set 5:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/723/","accounts_in_message":[],"_revision_number":5},{"id":"2a7f4849aa98dea7b13182a57f54a81885d539ed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 10:26:46.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"4092c430bffa0bc81c912dba6cc1f94abad29722","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-29 11:11:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959105 : FAILURE in 40m 05s","accounts_in_message":[],"_revision_number":5},{"id":"725cd03f3f82e8c640f1edc04ae9954047b8a7db","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-29 11:11:32.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2938e37803e345468766fa2ee02c6625 : FAILURE in 39m 16s","accounts_in_message":[],"_revision_number":5},{"id":"f852e7adbe9948d5360135ffa4a787c6e99072a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 11:14:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70e454e6d7d244d7921d2ffaba4bddb0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a262153b40b42b48e7557ba9b12a6ea : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db436088506547e89307bfd9fec8c7ab : FAILURE in 11m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0bfa572daa14e97925368e4c3eb7e1f : FAILURE in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b91f9be5b33d459e80569e536a126e31 : SUCCESS in 7m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4405fddb47f846cea4d0b14c825a0482 : SUCCESS in 36m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/87d226d933bc4154aa09961d183b9940 : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f832e7a19f704d49a4748bd11ad3e187 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":5},{"id":"fed5eb45f52198f82d79ca437f5e15fc13a596ff","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-29 11:33:12.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ic654fe7be872345785c29914a12206d6e3c49a2a .","accounts_in_message":[],"_revision_number":5},{"id":"6af26e017f7279e71198b9fc3ecceb235cd25540","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-29 13:35:36.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e25af00d0ae145efb5e0f4f2f90feb67 : FAILURE in 43m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/69b87e7ccaec457abb5aeb6a8965f915 : RETRY_LIMIT in 14m 57s","accounts_in_message":[],"_revision_number":5},{"id":"61c08dfe4a2ae4f921a2ca367f5f153626abce4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-01-29 16:18:07.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"159ee0ded06e52ae5dd508c6202410b3dc612901","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 16:18:14.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/745/","accounts_in_message":[],"_revision_number":6},{"id":"290fe8ddfeab789021cdc42d5018e627870da7f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-29 16:18:18.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959105/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"9c29affef474d98e7c40760c00fa8f180885750c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 17:07:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/409586188adc4ada97459c175b9864b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/407150f98ba7474a8dd3bb89e7d28b4b : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5220693fcf447158f3d5014843d271c : SUCCESS in 12m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1608c5a2e4e94952b53076e36a9ec27c : FAILURE in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6208e8cf6c4306b0769fc69ea68e50 : SUCCESS in 14m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9991611478745bfbb0c3f01b7b9a997 : SUCCESS in 47m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/911746013d704695b9d70cf619eb6d9e : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3ff0b4b8aa1e40b9b461a83a2d71f13f : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":6},{"id":"cdc260eb09a347d88f3eb1c6a07c4b96f2f92335","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-29 18:04:33.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ic654fe7be872345785c29914a12206d6e3c49a2a .","accounts_in_message":[],"_revision_number":6},{"id":"f9e47e28629dce30c705c0525d33d9c9c641b359","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-01-29 18:36:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9dd4f6c20a1ccbea4cd14e2dc092bb3dbf6e0e9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 19:22:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fad87eae53b410db791e388adebbf49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a11da6b198d84e65937b3f72d70c8cbb : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1bf50308285461db257b2827ddfcbc6 : SUCCESS in 9m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0204e5f6fd144d65a669d33531b46a88 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb437ae0dc2b47b78f2b78eaa71e9ac4 : SUCCESS in 12m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bc3b0dbffdbf41e98734dcc23f9f3a21 : SUCCESS in 42m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c58dac9a08eb4e51ae864b9968b63664 : FAILURE in 14m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4f8b66eec7a24b6c957bd75790ac33c9 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":6},{"id":"621d2ab18be60d9bfec8cc1a6aac3eae25833214","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-29 19:29:32.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_01_2026_19_22_13\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"19d96475b6b02e18be327ecd7c6726be5c83e301","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-01-29 19:43:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"877358484c3c6e0729f333533c74ed50db239108","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 01:53:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/572d4e8ce37b4e45908b5ccb47bc1050\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/195153f5975741789e7bbb082a22f298 : FAILURE in 19m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/041d2c32d2504132bb8848bc723abfa5 : FAILURE in 16m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7d6026d463be49179018c284c4b6cf64 : RETRY_LIMIT in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d9a814bd7352aef83310d195413eab7daa90cf3e","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-30 02:36:56.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/4c1180a7e7e148aba04550bee8ba78e5 : FAILURE in 40m 25s","accounts_in_message":[],"_revision_number":6},{"id":"270d5f32c4377489105522c025a1e99a58002a2f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-30 03:21:46.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959105 : FAILURE in 38m 48s","accounts_in_message":[],"_revision_number":6},{"id":"0222f51b386705d03d509a667ddf80b6031b93c0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-30 11:37:37.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/60c7e3e0ee7d495c81151e50fd0c6dbd : SUCCESS in 45m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a6a5606451e64503b506dfc6c8652c42 : RETRY_LIMIT in 15m 06s","accounts_in_message":[],"_revision_number":6},{"id":"675633ed88b5aef6a8e32e8f0aab0b370c2f1383","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-30 14:12:15.000000000","message":"Patch Set 3:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_407/index.html : SUCCESS in 96m","accounts_in_message":[],"_revision_number":3},{"id":"7b6d1f7e6ef2262e9f5b1611ca2c0b9c4d72c03b","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-31 00:06:17.000000000","message":"Patch Set 4:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_413/index.html : SUCCESS in 107m","accounts_in_message":[],"_revision_number":4},{"id":"cc6c4859160f68e952a25ec5234dd56d9f227d9b","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-31 01:52:34.000000000","message":"Patch Set 5:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_414/index.html : SUCCESS in 106m","accounts_in_message":[],"_revision_number":5},{"id":"ee46c0c4b48f18affb6ad044c6c01223948278fd","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-01 11:14:45.000000000","message":"Patch Set 6:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_436/index.html : SUCCESS in 91m","accounts_in_message":[],"_revision_number":6},{"id":"2ff25b5f2cf6a136fd1d904341cf6b2d34f28d67","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-02-01 19:41:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6db7f3993258f4d83217e479e6c391083799562e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2026-02-05 21:44:52.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/959105/6/20260206023426/NEC-V-ISCSI : FAILURE in 3h 28m 10s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/959105/6/20260206023426/NEC-V-FC : FAILURE in 4h 10m 24s","accounts_in_message":[],"_revision_number":6},{"id":"aa443241e870d3ab136b269c2c9e3172358bc500","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-02-06 14:17:37.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/959105/6/20260206130047/NEC-ISCSI : FAILURE in 9h 58m 55s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/959105/6/20260206130047/NEC-FC : FAILURE in 10h 16m 47s","accounts_in_message":[],"_revision_number":6},{"id":"dcc340592cd4e76c6fe06eceb9c8d04b6ca9cdf6","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-02-17 19:01:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"162826379b52b625b5c7f025e328cf7a5e0ffb3c","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-04-07 05:04:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c3d565692a343c4d49a79a4388944dd41af318c5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-07 05:11:04.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6c4f26ea8df8405299cfbe6ef6fcddee : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ca0a44256392417d926fd97bd414401b : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":6},{"id":"e6c850c50caccde43668f27e8eac3b8508e92c77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 05:24:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef0f0e392db8424a99bcda052e52aebc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3b2e7bd3fa6411983d663e3f74e1d97 : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94600e5b0be943c487b493309d74c10d : FAILURE in 11m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b9cf14f926f54a97b44e430965f92f3b : FAILURE in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"669eba4bd014045a495c2aaa96f496fb3696a5c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 06:13:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a7d11c34e044531b9da9df11634afb3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eebef7d9c937434da552351bc406c8e1 : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0215afa1d935450eade1a99ed63122d4 : FAILURE in 11m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eceb8b9464164ce6a46034dff3db9589 : FAILURE in 10m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/641aef27c3ef40ea9fa748f87c4f8293 : SUCCESS in 13m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/708ddae118754bca9864dee2bc946459 : SUCCESS in 12m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/004e870491554791a48fc011a73051b3 : SUCCESS in 1h 03m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f90b5522e1b642e1bd12ba27550ce7cb : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a9601bf4efa14e81900a838fcc921906 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":6},{"id":"2ce1824033ca36c091f06415cbf9bcc52bca808a","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-08 16:47:03.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a462c2f18e4b0f4c9b92cc1ddf007db24523aed8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-04-17 21:05:50.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"26ee6b4e34a114a37a38b41da2a5aa3bf5fc9b88","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-04-17 21:08:53.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"862138cb8cee9f37049f7ac43590bdff95b0ed7e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-17 21:11:04.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d748c670c0b046599df8b28887b9e96e : RETRY_LIMIT in 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2f60628cc73d43f19dab8cc8fa26f7a5 : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":7},{"id":"a0a333dc65b1cc8805284e2af1adf092c8b3d1e9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-17 21:45:00.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ic654fe7be872345785c29914a12206d6e3c49a2a .","accounts_in_message":[],"_revision_number":7},{"id":"95d454549f9ac1ad78196bfeeb1996a4c0d9420d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-17 21:45:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959105 : FAILURE in 38m 39s","accounts_in_message":[],"_revision_number":7},{"id":"21a3df18512b471b69446c9b99056de9e652f77f","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-17 21:46:42.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hpe-fc-driver build/8723e1088b3c4519994d786282e48455 : SUCCESS in 40m 28s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/959105/etc/tempest_log.log","accounts_in_message":[],"_revision_number":7},{"id":"b1c9ffb02f74967fcd8e7c08875a42dc411d8553","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-17 21:50:16.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a01cb1bd8c204ea596644198920bac33 : SUCCESS in 43m 51s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959105/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959105/etc/tempest_log.log","accounts_in_message":[],"_revision_number":7},{"id":"50fe3aadd36c2d7ebc03b98b28ca415cb0feff2a","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-17 21:52:04.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/96a48495caac49b6b3d74cbb73db8cee : SUCCESS in 45m 51s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/959105/etc/tempest_log.log","accounts_in_message":[],"_revision_number":7},{"id":"e3f434ace5d9cf1315cfaef4992b022c1238338e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 21:59:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f778149102524a9891765c3a9fe2c3bd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3ee141c381504f799ffab7e35fa66d6b : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/63d41677d04c45cc854cb4d6ec7d3a30 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b43ffc340fbd452baadf6f12f13a1850 : SUCCESS in 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f9718de18e388b7027a18f8f8617cc42f8e3e1d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 22:17:08.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a5dc0eec49a4cecae5d19aa198866fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55f349e095e44efd9a7ce54a95d7f340 : FAILURE in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ff8a22cda644ff5a166d5ea6e8e6245 : SUCCESS in 9m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/136fbb65ae5944569a199a563666aa8f : SUCCESS in 8m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/764bd166ccec4f26878fc347e75b8167 : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12731f1865f24c64bcfe0ef5e1ae1af5 : SUCCESS in 6m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c0f4d3723cc2492d9245dbcbe1746b67 : SUCCESS in 1h 05m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/98ba50e2db2a4b9a97b50287aac25499 : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/955a054d17184e4eaadd06f6403572f5 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":7},{"id":"084485807e971854de605c9d0d257c36b01c0f02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-04-20 09:21:14.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"518be375057ee841ef7d28ef7cc1e60f304cfd01","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-20 09:26:12.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/44f9026fab114b589a8362607cd27d84 : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/43fb72601c7a4c7f9681acb61a0591bc : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":8},{"id":"86f4af7022b755c9f12d00b1defc1cc0174c2b25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 09:55:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6354798855dc468eba886f542941d490\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a6168375d1b41e199cd6e7fd7ac20eb : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cc0a6a51e1f249238ec7264f5e01adb6 : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a19fdcdcd67a43759ef7c72a11c01d6d : SUCCESS in 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b92239ddc9e84e312354fbf9b9ad97ecda7f2e91","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-20 10:00:52.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ic654fe7be872345785c29914a12206d6e3c49a2a .","accounts_in_message":[],"_revision_number":8},{"id":"15019034dcc719a301bdc2587bb3dfcbcfc394ce","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-20 10:03:09.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959105 : FAILURE in 40m 31s","accounts_in_message":[],"_revision_number":8},{"id":"b3e78fa55b64e667bae6fbbb9058e53e5aac6781","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-20 10:05:55.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/253167dfa6d847ceb6622496150b78b2 : SUCCESS in 44m 13s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959105/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959105/etc/tempest_log.log","accounts_in_message":[],"_revision_number":8},{"id":"3e707a0161c3c9f6a065eb5b1fefd1b9c009713b","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-20 10:06:56.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/1533de01bea2449185838a36ff844a88 : SUCCESS in 45m 11s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/959105/etc/tempest_log.log","accounts_in_message":[],"_revision_number":8},{"id":"23473f44d2853e61654cda1bdf0245af5c377668","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-20 10:08:31.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- hpe-fc-driver build/2f49e8b4adf1475ba8b6f14ff61b774e : SUCCESS in 46m 51s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/959105/etc/tempest_log.log","accounts_in_message":[],"_revision_number":8},{"id":"2b400d0f8e595cd9bb6436b6dbad3bb9644d512b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:31:25.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83cf8edb0f674e5585a6c4cdd046dd84\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580f38baab8b45cea929e7a641a20fda : SUCCESS in 7m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3bbb4459e6b042f487a7588723ca9c7c : SUCCESS in 12m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c07d5c75e264498aa24f2796b6a5a9b0 : SUCCESS in 10m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d46058124b3f4d6ca81bc5bfe21685d4 : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beef22699350448eac28e7949c564053 : SUCCESS in 14m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fede0f728334429494a1e5d1b1fc1780 : SUCCESS in 1h 04m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e1439ea5416a42558a191275b98de6b7 : TIMED_OUT in 41m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eb76693db10b4ded9cdf7b5b95e6b8bb : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":8},{"id":"a45b5052c854a1146ca81c3bb13306ce59520fde","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-20 10:31:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3895/.","accounts_in_message":[],"_revision_number":8},{"id":"e5307aabb4e32100c11f8798b9a590b41d8b55d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-20 10:41:18.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_04_2026_10_34_45\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"e10bac923c46cd40926d881936003dd9602eeef8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-20 10:45:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3719/.","accounts_in_message":[],"_revision_number":8},{"id":"17f2a618e418459a1abc96a658fe8bab12544fd9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-04-20 16:17:35.000000000","message":"Patch Set 7:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/1546/ : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"7215e00b447eb5ed8ec97a0e092e9a9b13480494","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-04-20 16:17:56.000000000","message":"Patch Set 8:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/1552/ : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"11bb9d404af21cc854d7097eb7bf69e7675d283f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-05-29 15:31:29.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"61948df8889f9368fc2df16a821f657aacd8c29c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-29 15:52:14.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest build/11d80e5ef83f4c558a74205bf780baf1 : RETRY_LIMIT in 17m 52s\n- cinder-storpool-tempest-iscsi-multipath build/c6dba4a39faa4905bc8154f11b216abf : RETRY_LIMIT in 19m 27s","accounts_in_message":[],"_revision_number":9},{"id":"24a251fa62315af8ca8f8492828e632ce2723f96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 16:21:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aa6716cf4294be3b12180c9bd9bd65b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db0362b2a804f1a8ad488ae9257c90f : SUCCESS in 7m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e74a40ccca4f4c5e99988489f6af0389 : SUCCESS in 10m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37e96fcc9237452796d36bdb52a152dd : SUCCESS in 10m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/43627ab0b92546c28b1c30971349c69f : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22243a34cc7549179c8ae1f2c9572b18 : SUCCESS in 12m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/80f4101757fa4f53a62989da194b01f4 : SUCCESS in 48m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/646e3bdffcb240e7839114f80a2ecc4b : SUCCESS in 9m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/20bc28b9c8ba4fefa06264c84af90a5a : FAILURE in 26m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3e2102f8afb54b2bbb0553095c930536 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":9},{"id":"5778d1f00e2b3bbcc7964de4589c1a21e2be4bf5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-29 16:32:32.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_05_2026_16_24_59\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"345108c110cab5fad002165e532252bfab1b6272","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-29 16:53:53.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/3056d4a7ad1b4bde9612af2984ca2791 : FAILURE in 23m 09s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/959105","accounts_in_message":[],"_revision_number":9},{"id":"417c6ba76dbda9a2f2fca59eac988b87a9ff1886","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-29 17:37:02.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hpe-fc-driver build/8029c026c2204a0db7baf9d5187815fc : SUCCESS in 38m 09s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/959105","accounts_in_message":[],"_revision_number":9},{"id":"25c6877e40f9529f62a61a108cf89e2fa3acbfb5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-29 18:08:13.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/b0912cf986dc4ba28f7e6be690b2fa1b : SUCCESS in 39m 44s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959105\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959105","accounts_in_message":[],"_revision_number":9},{"id":"85548bee37545302b73c83917e28be162623b555","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-29 19:55:51.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-05-959105-9-fc : FAILURE in 58m 35s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-05-959105-9-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"37358701231baf3c8f480e6e769d0fa84c9928a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 20:03:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b947146932384b08a466d5baefcb42f5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2aa77ff12ffc4beb860b59128c5cecb3 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0ec6d486d03d40a08a6e3676f3d0b26b : SUCCESS in 19m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4092a6f185da444aa4e7a3899ccaf9c9 : SUCCESS in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"53b677b7a23dff71cd1c62bead494264b3f7a539","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 20:18:40.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4109/.","accounts_in_message":[],"_revision_number":9},{"id":"a6d6dbdda22ff1a6106e8d7eded056adbfffb5ca","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 20:27:00.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4285/.","accounts_in_message":[],"_revision_number":9},{"id":"9ecb852c3768bcc330129b269139fc007c1d4ebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-06-18 19:40:28.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"f23fc2af910ff63585f7b939519f31635fb0def2","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-18 20:03:35.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/de7c9a503e894e6e93fa614b6d64f72f : FAILURE in 22m 35s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/959105","accounts_in_message":[],"_revision_number":10},{"id":"50d88fd5150ebc22d9f077948a0e722ab2552fba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 20:06:38.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa1c451ebbd342928a62c49c42ba66f7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4614d66a95bf467f95909cba96bb9ec5 : SUCCESS in 17m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aa06cae727a14024820f3e894beab8b1 : SUCCESS in 19m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dfb7fc44d9af4fb989d27df5b7594015 : SUCCESS in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"046de4e193732e65a6b4595f1a12d27ec54b8985","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-18 20:12:51.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ic654fe7be872345785c29914a12206d6e3c49a2a .","accounts_in_message":[],"_revision_number":10},{"id":"d8a10e799d67b934ea7ef0337d6fe0fddc51661b","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-18 20:15:48.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- hpe-fc-driver build/afc597f416bd4f4d980c7879714614cc : SUCCESS in 34m 52s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/959105","accounts_in_message":[],"_revision_number":10},{"id":"085147be3d0effc85eda9fcd59e299f182627b08","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 20:21:26.000000000","message":"Patch Set 10:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/509 : SUCCESS in 0h 40m 40s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/959105","accounts_in_message":[],"_revision_number":10},{"id":"98e57c91797638e5b1e2164e1e6094c628b9e553","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 20:21:56.000000000","message":"Patch Set 10:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/155 : SUCCESS in 0h 41m 7s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/959105","accounts_in_message":[],"_revision_number":10},{"id":"57fb6b37aa6162b61ad8712f7489b2813e79ced4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 21:10:37.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ee53426aee24e168043d9cf8d4cda6f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fa49e195e5b44419dba569e7d540db1 : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49a31b23d94f4006a97f890e165036ae : SUCCESS in 11m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b25f801f91d48918cf4cc87cdfdef7a : SUCCESS in 6m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a150592b9ef74d46bdc71c07a56edad1 : SUCCESS in 11m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e27bf4588ff46df95d38cf7f8dde62d : SUCCESS in 7m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df2eb789de2048a49ffb9bfa2369fef5 : SUCCESS in 1h 20m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b03f41e4f444971b9fc5385b24967f6 : SUCCESS in 7m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6e9a3f443290442faf5a4d1dfb998840 : SUCCESS in 16m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/64be888a230049b7b1ade7c69d09e5fd : SUCCESS in 6m 37s\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/48830a7ce7fa4534b4af8d1080900e21 : FAILURE in 19m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f204f277e4dbe079d19df1d00a25820b527cd25e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-18 21:21:36.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_06_2026_21_13_48\nLeave a comment with","accounts_in_message":[],"_revision_number":10},{"id":"de56d2ec6061199a8dcbea6bc226beb0dfa466ae","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 22:41:12.000000000","message":"Patch Set 10:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/241 : FAILURE in 3h 0m 21s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/959105","accounts_in_message":[],"_revision_number":10},{"id":"da2d1bffd904c488c93fdb63e8e60b4340a196b0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-18 23:48:05.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/fc2496e0ea9b4368939e94f91b35df48 : SUCCESS in 41m 06s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959105\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/959105","accounts_in_message":[],"_revision_number":10},{"id":"507793021ce0caec90e30ad0ac16c59e9fcdc5c1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-19 02:34:18.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest build/1e40215c136d492ca904143822415906 : FAILURE in 50m 43s\n- cinder-storpool-tempest-iscsi-multipath build/e6942326c769460db8def1d9eb153079 : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":10},{"id":"e04dbc53841bd15518b0b8372e5fac95175404e9","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-20 20:53:39.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-05-959105-10-fc : SUCCESS in 58m 49s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-05-959105-10-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"9126181834e7e309ea7beb1348901f1619fae43e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-06-26 18:21:16.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/959105/10/20260626212041/NEC-ISCSI : SUCCESS in 5h 34m 25s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/05/959105/10/20260626212041/NEC-FC : SUCCESS in 6h 00m 32s","accounts_in_message":[],"_revision_number":10},{"id":"0cf8db15a26901223d8d5b8b0945191c1d60c802","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-07-13 12:08:55.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"9b173957cdb585352345d48fdfaa3be2e7d20cb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 12:46:01.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ceccb18f42945bca99b0a46be43186c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/36941816941d4d28b23f660f52f8ba5f : NODE_FAILURE Node(set) request 3de8c99f362940c9af65df8a672fa9d8 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/58b71af39345478bb5a24adc73bfdcee : SUCCESS in 17m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/268d185a0d1142c4a5c57141ffc3bc32 : SUCCESS in 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4bfa9c402a447e55295ce032fa1f996a5aadf7f0","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-13 12:51:58.000000000","message":"Patch Set 11:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/860 : SUCCESS in 0h 42m 40s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/959105","accounts_in_message":[],"_revision_number":11},{"id":"24701ea4517225efca949da77d65fa3b9818f74f","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-13 12:52:25.000000000","message":"Patch Set 11:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/503 : SUCCESS in 0h 43m 10s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/959105","accounts_in_message":[],"_revision_number":11},{"id":"9024bda8615c4f002c5478b8d96d2df3feaea539","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-13 12:53:19.000000000","message":"Patch Set 11:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/591 : SUCCESS in 0h 44m 5s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/959105","accounts_in_message":[],"_revision_number":11},{"id":"e5ac6fb8d932c8981363030644b63e48bbd4b51e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-13 13:00:59.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/5caf22e9afbf4dcf98c8aa470959ca98 : SUCCESS in 49m 32s\n- cinder-storpool-tempest-iscsi-multipath build/ef70b316b701438bb9727e92940e4757 : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":11},{"id":"1ccd5c90717b73af2c4f4a0fd1f3a01de7aed563","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-13 13:03:14.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic654fe7be872345785c29914a12206d6e3c49a2a .","accounts_in_message":[],"_revision_number":11},{"id":"ec2788051620d3445996f89987d66d71c5c2a1e4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-13 13:22:27.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4782/.","accounts_in_message":[],"_revision_number":11},{"id":"05b13bdc2ed69b8c7c0fa34ede268c85af43a058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 13:30:33.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39508821e8df4ef383c34699a2acec80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/330c3b66a1a1458582f6f7ddbc35dade : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1f94a4369ac434e8e835307275b409f : SUCCESS in 10m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8cfc629064f47528d244eefc25ac106 : SUCCESS in 10m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4c520c9f91ff48259cc8c9bbe5f34486 : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf302fb490054cb8893ec651424bf48b : SUCCESS in 14m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d77821dd35fe403b839a35e7b366cc81 : SUCCESS in 1h 15m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/833482539ef240b29dcf9a08c7e28f05 : SUCCESS in 7m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d93c2ed692ea4501b0bd51876f62f006 : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/08f46f713175490db46a283b452d3066 : SUCCESS in 4m 40s\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/9fbb5c1bcae74c4ba0f8b29cedce4185 : FAILURE in 13m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8d660008fcf11c2d4c0fb6fa454964559b08a6c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-13 13:40:55.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_07_2026_13_34_02\nLeave a comment with","accounts_in_message":[],"_revision_number":11},{"id":"c00d41fbca989d1d1894714e646503c5e3dc65d9","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-13 22:57:37.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-05-959105-11-fc : SUCCESS in 1h 11m 24s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-05-959105-11-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"7c34ed767a805108806a7bfe4af6bf7c59514f6b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-14 07:12:59.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-959105-11/FC/ : FAILURE in 19h 3m 24s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-959105-11/iSCSI/ : UNSTABLE in 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"7e6bb6d67dbd6a247069d88eca9e2f82b9b20e66","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-15 10:03:21.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"23610533ab9ba89c18aac9305932caa7a4afb5dc","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-09 12:45:26.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7d1ddce77fd4b33e36f24ea1fcce04a17ff77588","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-09 14:49:11.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8c98612e73cda78a69d7f8e6aa034852dca284d5","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-09 18:25:14.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f367928d5ffe9993551d28ed33774c30823e56f1","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-10 10:07:24.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2a4dcb6b41dddcb76e86ac1cd7e2b9ad3599a9f7","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-10 15:48:52.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/9e2390505b754144877a7a82acc981ba\n\n- powermax-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/7d630e80fd1941d08ede99c1dab498e7 : FAILURE in 2h 54m 19s\n- powermax-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/a9e0235215de4cb4a194980eb36cbe20 : FAILURE in 2h 50m 02s\n- powermax-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/5833f6468a744cec920bebd52ba84bcc : FAILURE in 5h 40m 22s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"67a8f2f1a4c0b3f6415170e0fbc0b8380dbc5445","revisions":{"e7d2e32f893caa53a665bf1a131e53ed07367ec1":{"kind":"REWORK","_number":1,"created":"2025-09-02 05:27:38.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/05/959105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/959105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/959105/1"}}},"commit":{"parents":[{"commit":"7926029d07f7de68f68c7468a85c48aedc238b1f","subject":"Dell PowerMax: NVMe/TCP suuport (follow-up)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7926029d07f7de68f68c7468a85c48aedc238b1f"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-26 02:18:10.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-09-02 05:26:54.000000000","tz":0},"subject":"Dell PowerMax: Improve UT coverage","message":"Dell PowerMax: Improve UT coverage\n\nChange-Id: Ic654fe7be872345785c29914a12206d6e3c49a2a\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e7d2e32f893caa53a665bf1a131e53ed07367ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e7d2e32f893caa53a665bf1a131e53ed07367ec1"}]},"branch":"refs/heads/master"},"b1f9a930497752d5159a58eaa47a31ef6c8512a4":{"kind":"REWORK","_number":2,"created":"2025-09-02 07:55:57.000000000","uploader":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"ref":"refs/changes/05/959105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/959105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/959105/2"}}},"commit":{"parents":[{"commit":"7926029d07f7de68f68c7468a85c48aedc238b1f","subject":"Dell PowerMax: NVMe/TCP suuport (follow-up)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7926029d07f7de68f68c7468a85c48aedc238b1f"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-26 02:18:10.000000000","tz":0},"committer":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-09-02 07:55:21.000000000","tz":0},"subject":"Dell PowerMax: Improve UT coverage","message":"Dell PowerMax: Improve UT coverage\n\nThis patch updates the Cinder Dell PowerMax driver Unit Test coverage to achive 90%.\n\nChange-Id: Ic654fe7be872345785c29914a12206d6e3c49a2a\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1f9a930497752d5159a58eaa47a31ef6c8512a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1f9a930497752d5159a58eaa47a31ef6c8512a4"}]},"branch":"refs/heads/master"},"03e3b6a24a8266f615da54384d6b7777dc1dbc03":{"kind":"REWORK","_number":3,"created":"2026-01-28 20:16:15.000000000","uploader":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"ref":"refs/changes/05/959105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/959105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/959105/3"}}},"commit":{"parents":[{"commit":"a0c22c68c175eff7fb277352a367883ce25e59da","subject":"Merge \"fix: add synchronisation during mount\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0c22c68c175eff7fb277352a367883ce25e59da"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-26 02:18:10.000000000","tz":0},"committer":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-01-28 20:15:38.000000000","tz":0},"subject":"Dell PowerMax: Improve UT coverage","message":"Dell PowerMax: Improve UT coverage\n\nThis patch updates the Cinder Dell PowerMax driver Unit Test coverage to achive 90%.\n\nChange-Id: Ic654fe7be872345785c29914a12206d6e3c49a2a\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03e3b6a24a8266f615da54384d6b7777dc1dbc03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03e3b6a24a8266f615da54384d6b7777dc1dbc03"}]},"branch":"refs/heads/master"},"88876f1b0d93d178a431921d3713614f7638abd1":{"kind":"REWORK","_number":4,"created":"2026-01-29 09:17:37.000000000","uploader":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"ref":"refs/changes/05/959105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/959105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/959105/4"}}},"commit":{"parents":[{"commit":"a0c22c68c175eff7fb277352a367883ce25e59da","subject":"Merge \"fix: add synchronisation during mount\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0c22c68c175eff7fb277352a367883ce25e59da"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-26 02:18:10.000000000","tz":0},"committer":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-01-29 09:16:56.000000000","tz":0},"subject":"Dell PowerMax: Improve UT coverage","message":"Dell PowerMax: Improve UT coverage\n\nThis patch updates the Cinder Dell PowerMax driver Unit Test coverage to achive 90%.\n\nChange-Id: Ic654fe7be872345785c29914a12206d6e3c49a2a\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88876f1b0d93d178a431921d3713614f7638abd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88876f1b0d93d178a431921d3713614f7638abd1"}]},"branch":"refs/heads/master"},"3f452aba8ba4bee020a11ebdfd4e0ce64e2151e3":{"kind":"REWORK","_number":5,"created":"2026-01-29 10:26:35.000000000","uploader":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"ref":"refs/changes/05/959105/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/959105/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/959105/5"}}},"commit":{"parents":[{"commit":"a0c22c68c175eff7fb277352a367883ce25e59da","subject":"Merge \"fix: add synchronisation during mount\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0c22c68c175eff7fb277352a367883ce25e59da"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-26 02:18:10.000000000","tz":0},"committer":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-01-29 10:26:05.000000000","tz":0},"subject":"Dell PowerMax: Improve UT coverage","message":"Dell PowerMax: Improve UT coverage\n\nThis patch updates the Cinder Dell PowerMax driver Unit Test coverage to achive 90%.\n\nChange-Id: Ic654fe7be872345785c29914a12206d6e3c49a2a\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f452aba8ba4bee020a11ebdfd4e0ce64e2151e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f452aba8ba4bee020a11ebdfd4e0ce64e2151e3"}]},"branch":"refs/heads/master"},"dd667d8c41fcdf4b48b7f272105c6e83711ce21f":{"kind":"REWORK","_number":6,"created":"2026-01-29 16:18:07.000000000","uploader":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"ref":"refs/changes/05/959105/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/959105/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/959105/6"}}},"commit":{"parents":[{"commit":"a0c22c68c175eff7fb277352a367883ce25e59da","subject":"Merge \"fix: add synchronisation during mount\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0c22c68c175eff7fb277352a367883ce25e59da"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-26 02:18:10.000000000","tz":0},"committer":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-01-29 16:17:24.000000000","tz":0},"subject":"Dell PowerMax: Improve UT coverage","message":"Dell PowerMax: Improve UT coverage\n\nThis patch updates the Cinder Dell PowerMax driver Unit Test coverage to achive 90%.\n\nChange-Id: Ic654fe7be872345785c29914a12206d6e3c49a2a\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd667d8c41fcdf4b48b7f272105c6e83711ce21f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd667d8c41fcdf4b48b7f272105c6e83711ce21f"}]},"branch":"refs/heads/master"},"dea7d8e2732410fbbd74f90b3e183ad0b17944fe":{"kind":"REWORK","_number":7,"created":"2026-04-17 21:05:50.000000000","uploader":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"ref":"refs/changes/05/959105/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/959105/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/959105/7"}}},"commit":{"parents":[{"commit":"a0c22c68c175eff7fb277352a367883ce25e59da","subject":"Merge \"fix: add synchronisation during mount\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0c22c68c175eff7fb277352a367883ce25e59da"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-26 02:18:10.000000000","tz":0},"committer":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-04-17 21:05:29.000000000","tz":0},"subject":"Dell PowerMax: Improve UT coverage","message":"Dell PowerMax: Improve UT coverage\n\nThis patch updates the Cinder Dell PowerMax driver Unit Test coverage to achieve 90%.\n\nChange-Id: Ic654fe7be872345785c29914a12206d6e3c49a2a\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\nCo-Authored-By: Abhishek Gupta \u003cAbhishek_Gupta14@Dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dea7d8e2732410fbbd74f90b3e183ad0b17944fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dea7d8e2732410fbbd74f90b3e183ad0b17944fe"}]},"branch":"refs/heads/master"},"4de15f228ce4d8974d8262b9594c5ca62771699f":{"kind":"REWORK","_number":8,"created":"2026-04-20 09:21:14.000000000","uploader":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"ref":"refs/changes/05/959105/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/959105/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/959105/8"}}},"commit":{"parents":[{"commit":"a0c22c68c175eff7fb277352a367883ce25e59da","subject":"Merge \"fix: add synchronisation during mount\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0c22c68c175eff7fb277352a367883ce25e59da"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-26 02:18:10.000000000","tz":0},"committer":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-04-20 09:20:41.000000000","tz":0},"subject":"Dell PowerMax: Improve UT coverage","message":"Dell PowerMax: Improve UT coverage\n\nThis patch updates the Cinder Dell PowerMax driver Unit Test coverage to achieve 90%.\n\nChange-Id: Ic654fe7be872345785c29914a12206d6e3c49a2a\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\nCo-Authored-By: Abhishek Gupta \u003cAbhishek_Gupta14@Dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4de15f228ce4d8974d8262b9594c5ca62771699f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4de15f228ce4d8974d8262b9594c5ca62771699f"}]},"branch":"refs/heads/master"},"5645a96238589086027eb6790bfdaad3fd1004f3":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-05-29 15:31:29.000000000","uploader":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"ref":"refs/changes/05/959105/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/959105/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/959105/9"}}},"commit":{"parents":[{"commit":"2ef4fadbca5ed10390926e56bba309cf34b4d4a0","subject":"Merge \"Fix cascade delete for non-admin with deleted cross-project snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ef4fadbca5ed10390926e56bba309cf34b4d4a0"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-26 02:18:10.000000000","tz":0},"committer":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-05-29 15:31:29.000000000","tz":0},"subject":"Dell PowerMax: Improve UT coverage","message":"Dell PowerMax: Improve UT coverage\n\nThis patch updates the Cinder Dell PowerMax driver Unit Test coverage to achieve 90%.\n\nChange-Id: Ic654fe7be872345785c29914a12206d6e3c49a2a\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\nCo-Authored-By: Abhishek Gupta \u003cAbhishek_Gupta14@Dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5645a96238589086027eb6790bfdaad3fd1004f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5645a96238589086027eb6790bfdaad3fd1004f3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"4de15f228ce4d8974d8262b9594c5ca62771699f","theirs":"2ef4fadbca5ed10390926e56bba309cf34b4d4a0","contains_conflicts":false}},"e0055b021cb772e91973040d73f884905ff0ae51":{"kind":"REWORK","_number":10,"created":"2026-06-18 19:40:28.000000000","uploader":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"ref":"refs/changes/05/959105/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/959105/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/959105/10"}}},"commit":{"parents":[{"commit":"2ef4fadbca5ed10390926e56bba309cf34b4d4a0","subject":"Merge \"Fix cascade delete for non-admin with deleted cross-project snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ef4fadbca5ed10390926e56bba309cf34b4d4a0"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-26 02:18:10.000000000","tz":0},"committer":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-06-18 19:35:27.000000000","tz":0},"subject":"Dell PowerMax: Improve UT coverage","message":"Dell PowerMax: Improve UT coverage\n\nThis patch updates the Cinder Dell PowerMax driver Unit Test coverage to achieve 95%.\n\nChange-Id: Ic654fe7be872345785c29914a12206d6e3c49a2a\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\nCo-Authored-By: Abhishek Gupta \u003cAbhishek_Gupta14@Dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0055b021cb772e91973040d73f884905ff0ae51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0055b021cb772e91973040d73f884905ff0ae51"}]},"branch":"refs/heads/master"},"67a8f2f1a4c0b3f6415170e0fbc0b8380dbc5445":{"kind":"REWORK","_number":11,"created":"2026-07-13 12:08:55.000000000","uploader":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"ref":"refs/changes/05/959105/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/959105/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/959105/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/959105/11"}}},"commit":{"parents":[{"commit":"2ef4fadbca5ed10390926e56bba309cf34b4d4a0","subject":"Merge \"Fix cascade delete for non-admin with deleted cross-project snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ef4fadbca5ed10390926e56bba309cf34b4d4a0"}]}],"author":{"name":"cuiyeliu","email":"cuiye.liu@dell.com","date":"2025-08-26 02:18:10.000000000","tz":0},"committer":{"name":"siddhvrth","email":"siddharth.kumar1@dell.com","date":"2026-07-13 11:39:23.000000000","tz":0},"subject":"Dell PowerMax: Improve UT coverage","message":"Dell PowerMax: Improve UT coverage\n\nThis patch updates the Cinder Dell PowerMax driver Unit Test\ncoverage to achieve 99%.\n\nChange-Id: Ic654fe7be872345785c29914a12206d6e3c49a2a\nSigned-off-by: cuiyeliu \u003ccuiye.liu@dell.com\u003e\nCo-Authored-By: Abhishek Gupta \u003cAbhishek_Gupta14@Dell.com\u003e\nCo-Authored-By: Siddharth Kumar \u003cSiddharth.kumar1@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67a8f2f1a4c0b3f6415170e0fbc0b8380dbc5445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67a8f2f1a4c0b3f6415170e0fbc0b8380dbc5445"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
