)]}'
{"id":"openstack%2Fcinder~959256","triplet_id":"openstack%2Fcinder~master~I544877f192375ab68a1bd8a547aa136c0137c8d5","project":"openstack/cinder","branch":"master","topic":"bug/2121812","attention_set":{},"removed_from_attention_set":{"35761":{"account":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"last_update":"2025-10-06 10:30:55.000000000","reason":"\u003cGERRIT_ACCOUNT_35761\u003e replied on the change","reason_account":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2025-09-12 11:33:05.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2025-11-01 09:36:59.000000000","reason":"Change was submitted"},"10366":{"account":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"last_update":"2025-11-01 09:36:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I544877f192375ab68a1bd8a547aa136c0137c8d5","subject":"[NetApp] Default node model to empty string","status":"MERGED","created":"2025-09-03 01:05:12.000000000","updated":"2026-07-29 13:58:50.000000000","submitted":"2025-11-01 09:36:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959256-bug/2121812","meta_rev_id":"34a3e5fe1d99e1289b2b1bb91b908ab72f4ae473","_number":959256,"virtual_id_number":959256,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-01 09:36:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"recommended":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"all":[{"value":1,"date":"2025-10-23 03:53:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-10-31 18:21:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-31 15:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-10-31 18:21:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"value":1,"date":"2025-10-31 16:40:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 01:05:23.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-09-03 01:06:39.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-03 01:35:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-03 02:10:36.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-09-03 02:29:20.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-03 02:39:29.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-09-03 02:53:36.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-03 03:30:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-03 09:26:50.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-09-04 10:02:23.000000000","updated_by":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"reviewer":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"state":"REVIEWER"},{"updated":"2025-09-05 15:36:49.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-09-11 10:37:31.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2025-09-11 12:06:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-11 13:56:33.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-09-11 13:57:18.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-09-11 13:57:56.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-09-11 13:58:34.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-10-23 07:12:56.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-10-31 15:40:12.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2025-10-31 16:40:55.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-11-01 11:31:57.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-07-29 13:58:50.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"CC"}],"messages":[{"id":"9a5ae85ea01b1e1d7dc9655c7e3b527033dd3706","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-03 01:05:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96e787fcd0277b41c5204f4c512d26b026c7c948","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 01:05:23.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/398/","accounts_in_message":[],"_revision_number":1},{"id":"eeb5518011b1312848a1a2565b121143b4312ca6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-03 01:06:39.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"9eb74f3f0449509fe6b4b2ea49d2d4a0c48573a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-03 01:09:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"69eca22e520728b01836ffa58eaf477e3ef9621f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-03 01:11:19.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"9542de88b5dacac612b5af9ad7735c8894de7a41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 01:35:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8da901382ef4deb84874dfc79617bb7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8d3564efed4a4a4995473881826dfb11 : SUCCESS in 18m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f63244b4c214c06b08fedf2c5df2979 : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad269bbfcb6c410788804a805ca6f06a : SUCCESS in 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0cc0bf1cd3cc5c8f20fd6f959540f789f92c018","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 02:00:43.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"a4eba891517917f261cb0901bf4cc8bd7ed03c7b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 02:00:44.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/399/","accounts_in_message":[],"_revision_number":2},{"id":"dce77d6797af24833e96b91a03e10f1277425531","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-03 02:10:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1652/.","accounts_in_message":[],"_revision_number":2},{"id":"b6b2675fc5cdadae5c1b66fd10675afad6655cd3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-03 02:16:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1787/.","accounts_in_message":[],"_revision_number":2},{"id":"5b48e46d978e25590134f1bf7cb76f52ab2ed04a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-03 02:29:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c6fee6131300442d8e22a67470ebbdda : SUCCESS in 45m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dd7f7c3ec1da4007921076e3b19f2070 : FAILURE in 48m 07s","accounts_in_message":[],"_revision_number":2},{"id":"4d150c64e5f3a34ccddcd99325ddf585ed84f0fc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-03 02:39:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959256 : FAILURE in 1h 25m 12s","accounts_in_message":[],"_revision_number":2},{"id":"d002f2251ecd47f60b7ce17dea01d19cdbe1c7c4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-03 02:53:36.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6573593b26c3d99f148892dd10011c86945074a1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-03 02:57:16.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"31e6133d2a8b236a9b8a1527217ce468aa7640a2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-03 03:00:25.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ad9a53c58f9b27247ba8b6c6ab8f1c6d67bb29db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 03:30:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b093bbf52b9d41f8836340cc5c4eaa49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ca43c71f47c43f3acb8e0ec47158db6 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/823922cc208b4a5a8d4f5fc02e76bf08 : SUCCESS in 7m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b1d66404345f48b3937a1705569d1a3d : SUCCESS in 11m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/860b60ca3e73402fb41bf3d7a020ba6a : TIMED_OUT in 40m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71c3b613a8c443f1a360909939e68f00 : SUCCESS in 14m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/4cad9109dc674e93ac0b14e69387f76e : SUCCESS in 28m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/27b2e1764243426ca74cc53d23ed9141 : SUCCESS in 1h 03m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3cd7a33e5af3419db0544f3ec5067891 : SUCCESS in 1h 52m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40a229132c744e34a6887fe8f4cbb7b1 : SUCCESS in 1h 00m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae5b5701ba7a4ddb8d6f3e16b9c18aec : SUCCESS in 16m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/45464c5ca3ed4413a889a4890d09c1c5 : SUCCESS in 22m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f2411014cc754e44ac0e12a069ea62a2 : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/19a37f341f3647cbaa5f23b27544b8d7 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b4b7f07c237a48b8933e16d10fc7df66 : SUCCESS in 6m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a0e518ba41a44bacaea6311bc9c38904 : SUCCESS in 9m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/89118ec58e9c4c3a8ab65729494641dd : FAILURE in 29m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0beedf069cda4a3c8c6787916db89a89 : SUCCESS in 5m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fa58d3a3db5347a3bc975b1e3b148bab : SUCCESS in 1h 44m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3ba51d0bbae946f5a344306f0d04cfff : FAILURE in 2h 13m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0ed45b836b4a48e8a37f21768758817c : SUCCESS in 1h 53m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e53eb14eddfd436183a9e600f64b09ea : FAILURE in 7m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fdf5cefd8a974362af9e883c0cb3522a : SUCCESS in 36m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c9f64fd4e6914d2cb6cc51d076cef7f2 : SUCCESS in 35m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cec035c6b20d405281c25ac20162dee9 : SUCCESS in 1h 46m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/20853d2088384b429cd7e6c3d012d203 : SUCCESS in 42m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/340f7b64608940e78af41866f974b871 : SUCCESS in 1h 18m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3df7035429bb456f9fadf49bb9e50a04 : FAILURE in 19m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1efa46a3fe3a4f81b8e1f6e4c406a508 : FAILURE in 2h 07m 05s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e7e42774e974189a660cffbd51eb3a2 : SUCCESS in 1h 33m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca1493c61f5a4daebd5a0a407c83fc51 : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":2},{"id":"41ed5741813906136b1fbb18a8a15b865a68a030","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-03 03:57:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d2051e38c2cc5955aa39ed7be7862d70bccaaee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 04:22:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f5a71282c6247adbccc382a3b73c7fe\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/083a5963d228443bacfaf2ea181970f0 : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb57036871b649e9b75213db5d33739a : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9fa7ffdc5e7b4cefaf60c4a724518b03 : SUCCESS in 23m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"429446aae01fc817e4a30c4a8c15dfd48969de8b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-03 05:25:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c72f403f781e444bb92a563e32244bc8 : SUCCESS in 53m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/53f4f64baa414893a8e440e343724d77 : FAILURE in 59m 47s","accounts_in_message":[],"_revision_number":2},{"id":"debf9813b12073a4f548bcad818460f981c14435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 06:09:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42ad9c6882714338b184d708734fc62c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ba477886a554e568f7c0de749f429a6 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4972c922f1144bd8b897b2bb6be568b6 : SUCCESS in 7m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/219e3ac8e7254987b48024b891b7634e : SUCCESS in 7m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27fb9bee36164678ab6b6cb1adf27587 : SUCCESS in 15m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb2d2fa36d14f90844b0c355eb855d2 : SUCCESS in 11m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/74708c43805a4a9e9b332dcfc5d14064 : SUCCESS in 43m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b051d82778c641c5a30b9179ac14ae66 : SUCCESS in 33m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c401c0f09b0435cb984ecf6e20c01f4 : SUCCESS in 1h 28m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cddde2c1b2e443aa1d9eb55da1d48d6 : SUCCESS in 45m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a98337ec0e6445e08fd02a20d3667f1f : SUCCESS in 15m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/335c77d587c34c39968ad21cc3b6e8bd : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7a605db125cb4072898e5a801b38bf61 : SUCCESS in 5m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30b504020f9b4950a8a00093ee6d50d2 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/07f752e7e2bd453cb4ab86e2146c7ec3 : SUCCESS in 6m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aea72872c7774c1ca4eddf5f3749ea58 : SUCCESS in 8m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a5fc8dcff4db45bf881ec2aa7c8757ad : FAILURE in 19m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9940907abaeb49b7beeea2bb30e5984a : SUCCESS in 5m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0ab31ebe447e425eaaef40a03bbd07ab : SUCCESS in 1h 19m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e635077ecafe4dc0844cc0ed2c3fd8d5 : FAILURE in 2h 05m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/37caeb742d674a279b8660ba87091412 : SUCCESS in 1h 57m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cdd88a72d20e4b32931b8cffc588cf44 : FAILURE in 18m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a3b7e377e5ff4228a01f73bb0cfe6eb8 : SUCCESS in 27m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c800dd4b81bb4f37878de72e1aef5b47 : SUCCESS in 1h 06m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7cab273762544385ae8a3f4afb89acdb : SUCCESS in 1h 31m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fca2cb65d2bf4101b454c7d0d6697beb : SUCCESS in 40m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cada5ca355744e06a3bb2f286499ef4c : SUCCESS in 1h 29m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4f9739f9bfd148d48e70d40dfc1d1082 : FAILURE in 18m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f295350b5cbd4b13bc9feeb0150ff0c5 : SUCCESS in 1h 40m 00s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d3bdedd11914477b81014fff9607ac1 : SUCCESS in 1h 33m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cef2618484a54506b10a9407a2b3d6d5 : SUCCESS in 30m 03s","accounts_in_message":[],"_revision_number":2},{"id":"c1b688ef52b3ecebc77da205164c9c7f4fb44683","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-03 06:53:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb8a2dedc3b3f7cf079eee11d10208667fcfe613","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-09-03 09:26:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/2/upstream-check/cinder-tempest-plugin-ontap-fcp/a191b42/ : FAILURE in 2h 32m 34s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/01e4680/ : SUCCESS in 46m 06s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/feaa0eb/ : SUCCESS in 1h 02m 37s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b69b77de3b6254b248595e85afb174824e099c1d","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2025-09-04 10:02:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1af06fd5ce43831e9f210854e1507d1563f1ab42","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-09-05 15:36:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43631d2997089ae99d3d29b4c59f09510dbde86c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-06 02:53:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f9e68f86b7a35479ee70c6ea3a65c851413faf83","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-06 02:53:22.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/483/","accounts_in_message":[],"_revision_number":3},{"id":"598beddcdf9a9df6bec90a2380d9ddc9e2226be1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-06 02:54:38.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"67e7969918f9d44dc0c1b45f5261f2e53bc81fac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-06 03:14:50.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"4f2dec55e1d40485ac72a1957cd68093cef8101d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-06 03:43:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bff25fb7c3cb4277b5abc31946772e5f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e874743c2a03404aab0d8af7e914b0fd : SUCCESS in 24m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a86e772e9a04f0dadd5738d9b3495e2 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9d0f269f231c40f098c760e2c77e3166 : SUCCESS in 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4a348042eafb563c5fbe123cb0bee4dd4680ec9e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-06 03:52:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1704/.","accounts_in_message":[],"_revision_number":3},{"id":"0101b0644332bb275e3ba598d5c3d7220d6d9bb2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-06 03:54:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1840/.","accounts_in_message":[],"_revision_number":3},{"id":"bb7a6e2e0e81e5071abc8898dcbec78dde51bf01","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-06 04:24:44.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959256 : FAILURE in 1h 29m 50s","accounts_in_message":[],"_revision_number":3},{"id":"4a1b9d38d3ca780d5e7b0a29f8478a377b2d13e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-06 05:10:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81c248b5ae7644f3a6b5a14ec9da4aca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/049e01db13794f9398fdb25075c2f79c : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/414877444e024d71a1949733896f3f40 : SUCCESS in 10m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ed199ef2d654d10828e38dc770683bb : SUCCESS in 10m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10b2ba01be2548faad9cf5835fd8457d : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc36a3c9774c41adb8ed6e72ce5a625e : SUCCESS in 17m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/ad41db2e1e4e4cb2836b440e260d6a30 : SUCCESS in 1h 00m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a85f4ed1b1a64379a38251f7132ab27f : SUCCESS in 57m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a941811ca4fa48b3ad43bf51883f7ed6 : SUCCESS in 1h 46m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74af16c9550a4e5bbfac3188f9b0e7ae : SUCCESS in 50m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72cea966cefe4f25bb66dea306efe72c : SUCCESS in 16m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d8c3c0a5938436b8bcb769e5b35728f : SUCCESS in 22m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d139186783de43bcb56979c66a912ccc : SUCCESS in 6m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/74ddab5544ca4f938cb64767baf9a396 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/089061972c954358a78450dfc4eeca82 : SUCCESS in 7m 18s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e234ba932abb4b84ab136649d6e63179 : SUCCESS in 10m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/052017f3f8cb42638d456f8f68215bf2 : FAILURE in 36m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/073b6536673f4748ac5e7a425c20942f : SUCCESS in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0a7bdfc8ab834d6f9190cd6d5ac602db : SUCCESS in 1h 36m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/838981ec1110445e84fc5f07becab716 : FAILURE in 2h 14m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/62a3dc4cc8934cd2ac5e6bf72cbb6b5a : SUCCESS in 1h 30m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/680000a439e54ac0900ab9aea926bfb9 : FAILURE in 11m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/14bbef472dc74f8b8945a6bd340b4624 : SUCCESS in 26m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bdda3e3711684aa59359dc2aeb3acc41 : SUCCESS in 1h 10m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3ee5bb816ffa47bcbd444be5c69d03ef : SUCCESS in 1h 38m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/95f3bb8f56cf46e18b22bf2cf3aea58a : SUCCESS in 44m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a04e11b8407547c7b5cae004d1036f69 : SUCCESS in 1h 31m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9ef501216f3c4d04b2cd28d513dcdec5 : FAILURE in 19m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4d9252e81854b638c850843b9b29ee9 : SUCCESS in 1h 42m 09s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3aeb6b48eeb4ba19ef76eea1562d578 : SUCCESS in 1h 24m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/012d69f24a44499a928c821dbb925afb : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":3},{"id":"2027f058be4dafc0d37a34536140ba79ab024c91","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-06 06:23:00.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/96721137b6c444fc8114bf0f11811d05 : SUCCESS in 1h 44m 18s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c3a4fbe38f644104b1ffdcc11aad6746 : FAILURE in 1h 50m 17s","accounts_in_message":[],"_revision_number":3},{"id":"ba10ea084804d73adc176002f9f77c26d85f1797","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-06 12:28:38.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":3},{"id":"b8f9e534938209a10a0c8f958d079e5f6c769de8","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-08 01:06:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"11d68898ecbe7cc6983a67fc87a317b7b7398f95","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-09-08 03:27:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/3/upstream-check/cinder-tempest-plugin-ontap-fcp/ac21426/ : FAILURE in 2h 04m 13s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/f79fea6/ : SUCCESS in 1h 13m 16s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8cf591d/ : SUCCESS in 1h 05m 53s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"17b090e106dd1425b2dfde7535c02a367b7de607","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-09-11 10:41:09.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"83f86daa4220e58b8f1566b75aaf22be94099e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-11 11:00:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"51842354002030aca70b9841f4b8e8e23a463b23","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 11:00:40.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/552/","accounts_in_message":[],"_revision_number":4},{"id":"7fad113a161be36149728254e431d452badfdbe1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-11 11:01:56.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"16ed504f93802eacbeb9b2d6e3f0b15469fb87a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-11 11:03:58.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c62da0c21acecab3cef4e4d4bae5b446f3d2c01a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-11 11:05:25.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"10eb994c3337983ad0737ef30a21eab0c78e4698","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-11 11:12:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bbfc11796927cdfd28fb9437b895eecd4ff1b77b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 11:19:19.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"ac6c781cc6fcc1c19c999b6fedb96549a88efa17","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 11:19:23.000000000","message":"Patch Set 5:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/553/","accounts_in_message":[],"_revision_number":5},{"id":"e977f5162ba0524cc119202f9b292d26040795bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 11:30:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c2d24ea9521472688e58b1082ab0608\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/17ed6f1a0d1e4907a56d7da3d1998378 : SUCCESS in 21m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1e2da690f8e473babb0c1ba58dc528c : SUCCESS in 23m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0f70970954cf42989982b5ddca783779 : SUCCESS in 17m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"07740df5c8e03a2d0941841342e1f0e0654fc7c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 11:38:38.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/5/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"29992e18970d9cb5ef459a3ae7658424d8d4181a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-11 12:06:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/959256 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":5},{"id":"761ce6bc4ff27b9bdf83ef2c04d25e68354d707a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-11 12:12:17.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/I544877f192375ab68a1bd8a547aa136c0137c8d5 .","accounts_in_message":[],"_revision_number":5},{"id":"ab8e0dea5d40249ec550f883af9d969b20154690","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 12:35:06.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1896/.","accounts_in_message":[],"_revision_number":4},{"id":"8db5b497ecee5d23bd6e6126fdaa91a580c630b7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-11 12:44:37.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959256 : FAILURE in 1h 13m 09s","accounts_in_message":[],"_revision_number":5},{"id":"62c3796a4e982c6afbbad6cc14e60a82882ac533","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-11 13:07:58.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/I544877f192375ab68a1bd8a547aa136c0137c8d5 .","accounts_in_message":[],"_revision_number":4},{"id":"47f181fb6f70c6622d8cf1e06c4615ce5336b0b4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-11 13:27:30.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/acb8d1f57fa8439b80cfd71fe5fa9206 : SUCCESS in 53m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ed4d2a85769d4568acce771ebb1141c8 : FAILURE in 53m 05s","accounts_in_message":[],"_revision_number":5},{"id":"affb12331bcf047b9dd8fedf9257461565002c71","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 13:38:52.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1897/.","accounts_in_message":[],"_revision_number":5},{"id":"e386ba2043e5877acdcec437872efe745951d968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 13:55:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08960955d36e49a5855776ed1cebcce9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9624c4d92384c8daf055a3b8a69770a : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/165add9081a743aa8c8f0056e209852d : SUCCESS in 12m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89988d1e1b4245599cea1d3f1490d721 : SUCCESS in 11m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5adb0c166b9b4656806af8061b8bc919 : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29bc5978f534d0bbc8aec581af34f55 : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/cbe913996e2648188431757dda1d6a89 : SUCCESS in 58m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9622cc19293c47ad92b6b20812727622 : SUCCESS in 59m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/14e7bbbbb52742c6b0d9783b7c53a16f : SUCCESS in 1h 43m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/803a87a0564a4af48328af17e792f47c : SUCCESS in 1h 03m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80acf2078a114e05affde4d57481c939 : SUCCESS in 11m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d3dcde059a1a48f7bb8ae7639e6706d8 : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8d87825f832d4128bb6a8a3f2231aa4c : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/164d9b6163694eb9a080bb26570c3cb6 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f8a65035f3f0446a860d60b1b799a082 : SUCCESS in 4m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/82a8362e0c48441a9bee69d551c9fcba : SUCCESS in 8m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4fb66d9a05ba4562bb6b40984f7244a0 : FAILURE in 31m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4aad74bebee94672b50ec249de23f62a : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f66d58d8641441daa2c8b779bad0d949 : SUCCESS in 1h 09m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cd6a38041ec949ef9f36ea17377b30d2 : FAILURE in 2h 46m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8b310e7d50434eb3b05c2a11a78f3e96 : SUCCESS in 1h 38m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/65797718dba349a5ace0f2d7a77a3f5e : FAILURE in 21m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e414727c97d9430282f6552adbb92e30 : SUCCESS in 37m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dd858dda41e942bbbcabd5f41f0b9fa0 : SUCCESS in 1h 13m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2b103d3f31de4581bba899c04a9ee0a1 : SUCCESS in 1h 30m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/31fdbbc09aba4401aab69fe5a234c961 : SUCCESS in 38m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9ccd87135fd5445b80452fc03d3f2e01 : SUCCESS in 1h 49m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/23ffcb6b7b38459b9721228b0f3e6a8d : FAILURE in 11m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/306e358a36214787b77ec4f654be8f24 : SUCCESS in 1h 36m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d70c7c6add424d509c7958f86bac3bd2 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":5},{"id":"935946e0160f9e64e8419d00149aba547a186f2d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-09-11 13:56:33.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"7c8f29053574f6a107d7c90504b79a5cb6457747","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-09-11 13:57:18.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"fc11c389d268076a60697f4beb4ca733ee42ffaf","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-09-11 13:57:56.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"ee1cdf84e41340f75127599b35c0bfb1f538ccec","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-09-11 13:58:34.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"66c5529287deaa295591362de75cce9ca69f78fc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 15:05:36.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1763/.","accounts_in_message":[],"_revision_number":5},{"id":"ff5a9ce2c00ff6c6f0e30e5f4a990096f2123b3f","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-11 15:50:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8e15d0b8b6412b6a2af0d87b66d529c8f915cbc0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-09-11 18:42:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/5/upstream-check/cinder-tempest-plugin-ontap-fcp/85cd41c/ : FAILURE in 2h 03m 25s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/1c34fbc/ : FAILURE in 1h 13m 59s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/bfc3136/ : FAILURE in 1h 05m 02s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"225901a734feb0fd145a8daa99353d36f2b37870","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-11 23:37:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9aa87e476401ee77cdca63688081c2ec11e1bd9b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-09-12 02:25:43.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/5/upstream-check/cinder-tempest-plugin-ontap-fcp/e7927f1/ : FAILURE in 2h 40m 00s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/0e7d6de/ : FAILURE in 1h 26m 11s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f9e4274/ : FAILURE in 1h 20m 43s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"6cdbdca93c1daa095c4c3856d327bb2e62998b80","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-12 03:10:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"644d5ca939600aadf256fd54723587a66966949a","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-09-12 11:33:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cb55754f71174972712705c953a48b7a6c879709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-12 11:36:46.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"2f645ae7ffae8c19bbe1b98aea035e066a4989d0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-12 11:36:54.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/575/","accounts_in_message":[],"_revision_number":6},{"id":"2bc375996d5cf689e5eda56f3e27ce61da325dc5","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-09-12 11:37:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"667384ef09dc0cdf1e35f40a5ca03b913ccaec3c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-12 11:38:18.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"6d2097f1971a849af271b5a2ef9645215413823e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-12 11:55:09.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959256/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"32a553046b5d6e1a67742899b3ab3dee67df51e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 13:02:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/374e91999be64dc488d3dc9b2bbe6c9e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c23a170c18c44d799b2ce48d5c9ae922 : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2500a0eff23e4962b4054c1a6dde8ea7 : SUCCESS in 27m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8b93e9810a34438e8eb6dc7db08078ec : SUCCESS in 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1765dd227ea36f1f3c6f08a85946c0e7378227aa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-12 13:24:02.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/959256 : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":6},{"id":"29eb86406bb077139019764720c363c5384ed7bb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-12 13:36:42.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/I544877f192375ab68a1bd8a547aa136c0137c8d5 .","accounts_in_message":[],"_revision_number":6},{"id":"be95fe9718cc35d1ad4c8fc4f8b73c2e5b846fb0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-12 13:56:59.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1920/.","accounts_in_message":[],"_revision_number":6},{"id":"695ddad0fa34754cd6d51fd215575bf4215b5be5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 14:09:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51d88ade3c184ff59c00afcc4dcff37f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d0afd7d26342d3bee3369e45f54a30 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0077a766fa0943ec8d6efd5ed09107fc : SUCCESS in 8m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ae73e925c3348f99c77bc4b98b4b9d6 : SUCCESS in 8m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f10bf102acb643828d0f2efd2f4d9d28 : SUCCESS in 15m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd774e0a5a24d1580d365968a4c5ae9 : SUCCESS in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/ffdcca6dbdc64ad0b9ac4d5b11f67df5 : SUCCESS in 1h 07m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4d67e4a4760748ba95ab439b3ad290c7 : SUCCESS in 1h 07m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47b5b4fdaa8340c2b8ab19b92f9719c5 : SUCCESS in 1h 17m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/644fc9757c5f436eb1e9fcda9204571a : SUCCESS in 50m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59fb81295326458a8512e01bb7ae9c74 : SUCCESS in 10m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/07ad852dbed04d939e4874c68976472d : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/43af0d557fa542ba936c72180e0e90bd : SUCCESS in 5m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd76ef75a6d9431ebb7e253539541b50 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/78a24e3697f54d798bbd2f0a281b1264 : SUCCESS in 5m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/81f41278afe549cba4275c809e7d1ac7 : SUCCESS in 9m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/00d8ab69b2d3497796cadb5072f8f3f1 : FAILURE in 28m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0b4141804b8c4b1abacd63c286b64122 : SUCCESS in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/47cc795475eb4c199ca46ad133834866 : SUCCESS in 1h 48m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c65fdccdd2f74b0abb58ed3118f89dab : FAILURE in 2h 25m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/62c8c557fa7b4f80b9dea2d07d8cd934 : SUCCESS in 1h 20m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cb7607cb540140988ea1055d0bd00128 : FAILURE in 12m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c72dc9091f36474b88014f5faeae0829 : SUCCESS in 25m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f3243f83f707442aa3aa3874560d88fe : SUCCESS in 1h 09m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bea3ea60230a48ee9e87a092225b1150 : SUCCESS in 1h 16m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1210f12d215a47989585b4fe1b9d80f2 : SUCCESS in 35m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e39ec1058c0e480a98382038450ffccf : SUCCESS in 1h 40m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/01416bc602b4470a9a122d35711a7950 : FAILURE in 19m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/22c7b37d78024c19bef4a7ffec2b0567 : SUCCESS in 1h 35m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1dc7b1e81de47db882109eb467aed6b : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":6},{"id":"f3c23efae5fe9db5aa6a6c800f64a4f81bd5253c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-09-12 14:13:15.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/6/upstream-check/cinder-tempest-plugin-ontap-fcp/f351c00/ : FAILURE in 2h 34m 20s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/b092383/ : SUCCESS in 58m 01s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8bf1056/ : SUCCESS in 48m 48s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"5e0a53a656052a1195d1e6630c94a326f68995b4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-12 14:21:25.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959256 : FAILURE in 1h 26m 02s","accounts_in_message":[],"_revision_number":6},{"id":"fbec349b55868e350bf630347dfeb8bb9c306f34","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-12 15:04:35.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1786/.","accounts_in_message":[],"_revision_number":6},{"id":"31535484a38a3449d26f7bee76b0a8c3ffaf0934","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-15 08:00:36.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9e10ee8d771448be8ba922211a99c7ac : FAILURE in 45m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b96acf4b97404f07a4c65e341a21ddff : FAILURE in 52m 38s","accounts_in_message":[],"_revision_number":6},{"id":"c863de99e78c9c7f5cb02cee2ca3daf50c0efa79","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2025-10-06 10:30:55.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"a7dc9e3d5fc339b1689e78c09580f5b48adcc934","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-10-23 03:53:54.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"97f96587ba837746541343c48ad9648a2a836969","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-10-23 03:54:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"23663b1eff6cb8e09c6adca8f729088593f5d877","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-23 03:55:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"a149f739342ca51651866d4b174e5c9b075a651c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 04:28:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/747ec6f1e3714c15822a2959529f527b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f992a5bec00b4f7e85b5fddb2ab93de9 : SUCCESS in 24m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b9b8bcb907b14037944b7a008711a729 : SUCCESS in 26m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0251270dba4148498dc0f3648812ecfd : SUCCESS in 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5e65f55d5fabd005442549c661bb094850c19762","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-23 04:55:34.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I544877f192375ab68a1bd8a547aa136c0137c8d5 .","accounts_in_message":[],"_revision_number":7},{"id":"b1bbc2ab2757e1fa41404d4c87c7f46d9d29c5f1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-10-23 06:12:24.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/7/upstream-check/cinder-tempest-plugin-ontap-fcp/55154a7/ : FAILURE in 2h 03m 17s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/c12a5a8/ : SUCCESS in 1h 14m 36s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/959256/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/cd4fa7d/ : SUCCESS in 1h 01m 13s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"f6bfd7294f33399da9aca07e1274abae16136780","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-23 06:25:47.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/01047baf55e04d1db779260239a22a0c : RETRY_LIMIT in 16m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/189f7bdf692b4d3996779efebe8ceb7a : RETRY_LIMIT in 14m 14s","accounts_in_message":[],"_revision_number":7},{"id":"b9d2b52ee00a2932701dbb8aa04e4fca3673ee16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 07:05:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/666b84f075964ca0be11069392d55bb3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2e1227864843e1bfb58c758e33c448 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5955a83cf6ee4b228f94eb46ce0b47cc : SUCCESS in 11m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fdcd56a997649b792a3e3097d30c7d2 : SUCCESS in 15m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/839e48621c5b453a9e36577081e3cce5 : SUCCESS in 14m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/3a95ced5f68e4542bb47bbd2c8f8b06d : SUCCESS in 1h 04m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abf6acce5d1544a2803b1b1a1010071d : SUCCESS in 59m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc8156cd345e41c5bb313e29e85fd4b3 : SUCCESS in 1h 25m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/804e2182d5994d53b5c05b4b916fe35d : SUCCESS in 58m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46e02ebefd8e438090f0774628534902 : SUCCESS in 16m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f5d7f4f37a9d4eb0861cc26a4dcaabd7 : SUCCESS in 20m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/97341d7b988a46ee808ec8d869f665da : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/758a70c7d3204d75b969b7409e5db793 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c51f9d3871a94d17a2b7bc8b2413dbab : SUCCESS in 5m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d28c16b9a71a4d5bb5930c93472d33af : SUCCESS in 8m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fdb443894f5f403486318751e4ca363c : FAILURE in 34m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f13f2a37b654422da3bb5826b727c0b3 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0a16db1a9fab4cd683eaa0b4d6086148 : SUCCESS in 1h 41m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3df10585cdb74d159456a74b40a3b396 : FAILURE in 2h 37m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a8aa885c855f4821aaf3bef182fdfc5a : SUCCESS in 2h 13m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b933ec7a0d864d28a09a8969d6a02d77 : SUCCESS in 1h 56m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/587aff0c244543fa9cea209079da7338 : SUCCESS in 22m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/05b98712e8eb40d783cde81d075fe875 : SUCCESS in 35m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eb1344a8d3e245d692cf4297f5deaca0 : SUCCESS in 1h 28m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/feff36d3fd6741ecbdc7b9d83f6c14fb : SUCCESS in 26m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3f00554ef379492584f95f01e652619a : SUCCESS in 2h 05m 58s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d4713dac4a5841ab8c08664bf5041b74 : POST_FAILURE in 2h 58m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8533dc8651c4effb721a4e69d187937 : SUCCESS in 1h 37m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/177850d0ae57483db8f2335538c543e3 : SUCCESS in 46m 23s","accounts_in_message":[],"_revision_number":7},{"id":"a400fdf30d979eefd4f9477af48c53d65a35cc5b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-23 07:12:56.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_10_2025_07_06_06\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"474614efda2b2ce07732ef6e33ca955de7c5f40b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-23 07:48:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2109/.","accounts_in_message":[],"_revision_number":7},{"id":"b3e62c93a8873353e4dbf637b796c2c1c703c13b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-23 08:38:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2223/.","accounts_in_message":[],"_revision_number":7},{"id":"4ac4f9113908d50fa5be0886e7ea9942f11297a3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-23 08:54:58.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959256 : FAILURE in 1h 11m 47s","accounts_in_message":[],"_revision_number":7},{"id":"0e2abb8c4ce191b25e3447b1e5ba820d56f2c1a2","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-10-31 15:40:12.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"449082d13e7fb92003391cb6958a0d33c5be2c49","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-31 16:40:55.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"831e024409d8b77432cbc11a51dd57c3c747ba3d","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-10-31 18:21:11.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"04eb0ee402de954664787ae4f058d13d49f820e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 18:22:12.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e1c674c3fb409d3aeef95912ddf2e6a5162337ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 21:05:03.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/323306321b3f4f0280f67781e0ee4e46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed251130f534d77a5e8b0c43449358e : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b2276c3597c46058365f15aad8e7cf8 : SUCCESS in 12m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f33e1528548045eaae2a8e610d457fe0 : FAILURE in 15m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8825589f0734e4a9229de36be601d35 : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/94d48d7d5b714f899754bd82721133ed : SUCCESS in 53m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/32b7890ed8034e42b9a1a350c918f2f5 : SUCCESS in 56m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f6b390ec17a47fdab1690d07ceb5692 : SUCCESS in 1h 47m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7879b24b14444d6a92ff092cfe87c8e3 : SUCCESS in 59m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3417fbf5d70e4036aaa76a4cb08cb74c : SUCCESS in 17m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/03e7acafd3a944058adeeb114f1ae379 : SUCCESS in 38m 05s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/681f0d392ca3486dbe5ef069867a6c57 : SUCCESS in 1h 51m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d354192d91454b33bbff8b02a756a465 : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":7},{"id":"280894b027292d9a072e69d2fe5d110db7962e50","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2025-11-01 04:43:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f698eda49414defba229b98cc1da920246ea573b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 05:05:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/803c0dca5635419cb836085db234bc6b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f477a21d564d431a966d8761678d62b7 : SUCCESS in 19m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/da36d7c84e384f618ac4db5703dd8b89 : SUCCESS in 19m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da2ca67c7494437b96ca1628319b3b1b : SUCCESS in 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"eb6cfcf8dccacfa0c417a0e38416b914daf221b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 07:39:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfef1753b91f46318d056938af32be61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cebf87ae62643c2a80df551e3675a18 : SUCCESS in 7m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5e58a9baf7e4f5198851bc31c7549a3 : SUCCESS in 10m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a33223e9706f44e1b9e9d4c3371b12b8 : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b604df51baf4c8db7f37a635934aa7e : SUCCESS in 11m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/6c79652dfad54cfa97c04c0eb2acc971 : SUCCESS in 1h 07m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/872d1039b8bb42f98ad9780f4d80cb96 : SUCCESS in 41m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84c02b5e7d5f4b31b6255871cf549743 : SUCCESS in 1h 34m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/063a7e31da5442d4abe6403b57c93e1c : SUCCESS in 1h 01m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f9e9e3659c84697978b6955b85b8142 : SUCCESS in 13m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/16a429070e024c0481b3ad49e0c4580d : FAILURE in 28m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cc0a32a2dfa546ed8e068b9851e17e1c : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7c98b744123b48f287f9dfde5f71c299 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca986c42adab41de96d553f9ae276d14 : SUCCESS in 5m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/20406f762d834566aca7af63df7c7a03 : SUCCESS in 7m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ffd7b30b7dfd421c94e28d895952f938 : FAILURE in 18m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a2e756640fcb4155b2c1e1c6046dbba3 : SUCCESS in 3m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd53faf8370a424fb31a1e1613ee1a1c : SUCCESS in 1h 48m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1c44874d631a4e8caa726a0dd4b8a93e : FAILURE in 2h 51m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8401cc8172cf431d959eb128b1391fc4 : SUCCESS in 1h 40m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2b3ca37c3f3f4827b46f775fe89e1c14 : SUCCESS in 2h 17m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/50cd1291fcf24cd48c5d16300f60d9ff : SUCCESS in 39m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bcf653cc543d4efc8615471f93d3f262 : SUCCESS in 1h 11m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7d4ac66bee074b9291faacb975a4516e : SUCCESS in 1h 43m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0629c846d7fc4adf876096e569df0df3 : SUCCESS in 44m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/112ab6088d8f42df95f2905603d9ba7f : SUCCESS in 2h 04m 20s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/fc81226ada7b4b2595ae88e5fe5975ed : SUCCESS in 1h 45m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/55a2260358634e44ba3e7fa521358d0f : SUCCESS in 53m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df8a6c52777f4247b449106314c0973a : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":7},{"id":"4131b0f9236c13622ad7d41d94fb6463b276ddb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 07:39:57.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2fd8e480fe9c120512ea2af48a213382d7269203","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-11-01 07:45:58.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_11_2025_07_39_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"cc804583d376c2e59ac7e7720fe09fd7f9bc2771","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 09:36:59.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15b92f30a11c41a5b32a861f533f1b4b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee4b288269f489f95afefc2c05bf280 : SUCCESS in 7m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74c29ffd84bc4ebebcb15c101f633355 : SUCCESS in 10m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1d09f9841ec487c8bcdedc7e8c67bd9 : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c146315091a349db8c8557329363705e : SUCCESS in 13m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/eaf97f5966eb4731b565faeb8444341b : SUCCESS in 45m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/735eefcdcb904038a4b345eb573a2e17 : SUCCESS in 47m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca622dc98dfb4b7cae7258b3c1d0dca6 : SUCCESS in 1h 36m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04e410bceba34453883b6773bd0dea8a : SUCCESS in 1h 01m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb4af152cfda47ec9515c619bc0f0239 : SUCCESS in 15m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/88f51b7b275041b585af8aad8efc8980 : SUCCESS in 55m 15s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/942c36832211455bbbd0598c3ea01bbe : SUCCESS in 1h 53m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0a548669991492981bc4307288eff98 : SUCCESS in 44m 53s","accounts_in_message":[],"_revision_number":7},{"id":"e54e2e4d975bc4d5c4b7296a61d44fbfdbb3118e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 09:36:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"3d76d8458f0df708f1778ee24e52364fd512348e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 09:38:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e53faa6486e4aac857b64a1344257f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfe14fdcd6f84d058ba8d3ddab8b2c19 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcaeb78b29dd4da6a45c3e6b9e24a05f : SUCCESS in 48s","accounts_in_message":[],"_revision_number":7},{"id":"675408576d48ddfe919a28d6a6c719ad2d7156a9","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-01 11:31:57.000000000","message":"Patch Set 7:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-56-959256-7-fc: FAILURE\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-56-959256-7-iscsi: FAILURE\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":7},{"id":"d6223c86439d2ded8937480e738ec7ad3d870741","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-01 14:40:27.000000000","message":"Patch Set 7:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-56-959256-7-fc: SUCCESS\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-56-959256-7-iscsi: FAILURE\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":7},{"id":"becf19fe0d7dd5d7a11070b078fc489e479fc127","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-01 17:42:17.000000000","message":"Patch Set 7:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-56-959256-7-fc: SUCCESS\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-56-959256-7-iscsi: SUCCESS\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":7},{"id":"63c5eee5e40eb5ec59e1533325cfb45015cef02a","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-01 20:57:24.000000000","message":"Patch Set 7:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-56-959256-7-fc: SUCCESS\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-56-959256-7-iscsi: FAILURE\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":7},{"id":"78e01b2b69bf008a754f5b93402dd25524249eb8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-02 17:43:35.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fdf668306b624e71b0928a523c9c530f : SUCCESS in 1h 37m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/eed35fef3a9d427e89a54787d7dcb85a : FAILURE in 1h 53m 47s","accounts_in_message":[],"_revision_number":7},{"id":"34a3e5fe1d99e1289b2b1bb91b908ab72f4ae473","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-07-29 13:58:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"711c97a1208ecca5c3bd96880677aa62d10cf34c","revisions":{"002c83c894c7bbcd16b3fa58f1a27801e8c56cc1":{"kind":"REWORK","_number":1,"created":"2025-09-03 01:05:12.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/56/959256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/959256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/959256/1"}}},"commit":{"parents":[{"commit":"3b35adc2c9a82bec865a6df1d8d7122961fcf262","subject":"Merge \"Fix typo in release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3b35adc2c9a82bec865a6df1d8d7122961fcf262"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-03 01:00:33.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-03 01:05:07.000000000","tz":330},"subject":"[NetApp] Default node model to empty string","message":"[NetApp] Default node model to empty string\n\ncinder-volume service is not started if the netapp\ndoes not provide node model in system node information.\nDefault node model value to empty string if not\nprovided by netapp.\n\nCloses-Bug: #2121812\nChange-Id: I544877f192375ab68a1bd8a547aa136c0137c8d5\nSigned-off-by: Hemanth Nakkina \u003chemanth.nakkina@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/002c83c894c7bbcd16b3fa58f1a27801e8c56cc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/002c83c894c7bbcd16b3fa58f1a27801e8c56cc1"}]},"branch":"refs/heads/master"},"2882466ba5780490c733b845687a6c85904ef699":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-03 01:09:52.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/56/959256/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/959256/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/959256/2"}}},"commit":{"parents":[{"commit":"aa0ce9963ffb5dd180d2f7161400b4c345f2bdc2","subject":"Merge \"Delete image cache volumes when they\u0027re unusable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aa0ce9963ffb5dd180d2f7161400b4c345f2bdc2"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-03 01:00:33.000000000","tz":330},"committer":{"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","date":"2025-09-03 01:09:52.000000000","tz":0},"subject":"[NetApp] Default node model to empty string","message":"[NetApp] Default node model to empty string\n\ncinder-volume service is not started if the netapp\ndoes not provide node model in system node information.\nDefault node model value to empty string if not\nprovided by netapp.\n\nCloses-Bug: #2121812\nChange-Id: I544877f192375ab68a1bd8a547aa136c0137c8d5\nSigned-off-by: Hemanth Nakkina \u003chemanth.nakkina@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2882466ba5780490c733b845687a6c85904ef699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2882466ba5780490c733b845687a6c85904ef699"}]},"branch":"refs/heads/master","description":"Rebase"},"7e6fd4f20adca913c9f8c99239a678b2264c82ee":{"kind":"REWORK","_number":3,"created":"2025-09-06 02:53:11.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/56/959256/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/959256/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/959256/3"}}},"commit":{"parents":[{"commit":"446f2453d158a7b067338c484e7d2fa22007e3a2","subject":"Merge \"[Pure Storage] Cinder manage quota breach deletion fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/446f2453d158a7b067338c484e7d2fa22007e3a2"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-03 01:00:33.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-06 02:52:53.000000000","tz":330},"subject":"[NetApp] Default node model to empty string","message":"[NetApp] Default node model to empty string\n\nIf a netapp node is down or in a transient state,\nit will fail to report model and this causes the\ndriver to fail init.\n\nDefault node model value to empty string if not\nprovided by netapp.\n\nCloses-Bug: #2121812\nChange-Id: I544877f192375ab68a1bd8a547aa136c0137c8d5\nSigned-off-by: Hemanth Nakkina \u003chemanth.nakkina@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e6fd4f20adca913c9f8c99239a678b2264c82ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e6fd4f20adca913c9f8c99239a678b2264c82ee"}]},"branch":"refs/heads/master"},"7900e7b097e58ead7ed2724ed1090ceb345d9006":{"kind":"REWORK","_number":4,"created":"2025-09-11 11:00:28.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/56/959256/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/959256/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/959256/4"}}},"commit":{"parents":[{"commit":"446f2453d158a7b067338c484e7d2fa22007e3a2","subject":"Merge \"[Pure Storage] Cinder manage quota breach deletion fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/446f2453d158a7b067338c484e7d2fa22007e3a2"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-03 01:00:33.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-11 11:00:03.000000000","tz":330},"subject":"[NetApp] Default node model to empty string","message":"[NetApp] Default node model to empty string\n\nIf a netapp node is down or in a transient state,\nit will fail to report model and this causes the\ndriver to fail init.\n\nDefault node model value to empty string if not\nprovided by netapp.\n\nCloses-Bug: #2121812\nChange-Id: I544877f192375ab68a1bd8a547aa136c0137c8d5\nSigned-off-by: Hemanth Nakkina \u003chemanth.nakkina@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7900e7b097e58ead7ed2724ed1090ceb345d9006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7900e7b097e58ead7ed2724ed1090ceb345d9006"}]},"branch":"refs/heads/master"},"0e212da9287bee9d7598fc1c800c93ff7c588231":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-09-11 11:03:58.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/56/959256/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/959256/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/959256/5"}}},"commit":{"parents":[{"commit":"fe5861dc19db2b67845b83f50d8fd56a9b978ddf","subject":"Merge \"[Pure Storage] Consistent hostnames across replicated backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe5861dc19db2b67845b83f50d8fd56a9b978ddf"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-03 01:00:33.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-11 11:03:42.000000000","tz":330},"subject":"[NetApp] Default node model to empty string","message":"[NetApp] Default node model to empty string\n\nIf a netapp node is down or in a transient state,\nit will fail to report model and this causes the\ndriver to fail init.\n\nDefault node model value to empty string if not\nprovided by netapp.\n\nCloses-Bug: #2121812\nChange-Id: I544877f192375ab68a1bd8a547aa136c0137c8d5\nSigned-off-by: Hemanth Nakkina \u003chemanth.nakkina@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e212da9287bee9d7598fc1c800c93ff7c588231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e212da9287bee9d7598fc1c800c93ff7c588231"}]},"branch":"refs/heads/master"},"60e39e33cef40a1ecb00620c99830de0f4d925ec":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-09-12 11:36:46.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/56/959256/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/959256/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/959256/6"}}},"commit":{"parents":[{"commit":"fe5861dc19db2b67845b83f50d8fd56a9b978ddf","subject":"Merge \"[Pure Storage] Consistent hostnames across replicated backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe5861dc19db2b67845b83f50d8fd56a9b978ddf"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-03 01:00:33.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-12 11:35:10.000000000","tz":330},"subject":"[NetApp] Default node model to empty string","message":"[NetApp] Default node model to empty string\n\nIf a netapp node is down or in a transient state,\nit will fail to report model and this causes the\ndriver to fail init.\n\nDefault node model value to empty string if not\nprovided by netapp.\n\nCloses-Bug: #2121812\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/960562\n\nChange-Id: I544877f192375ab68a1bd8a547aa136c0137c8d5\nSigned-off-by: Hemanth Nakkina \u003chemanth.nakkina@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60e39e33cef40a1ecb00620c99830de0f4d925ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60e39e33cef40a1ecb00620c99830de0f4d925ec"}]},"branch":"refs/heads/master"},"711c97a1208ecca5c3bd96880677aa62d10cf34c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-23 03:53:54.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/56/959256/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/56/959256/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/56/959256/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/56/959256/7"}}},"commit":{"parents":[{"commit":"d02171164bdd702b12b59888b744d172f30d712d","subject":"Merge \"docs: starting iscsid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d02171164bdd702b12b59888b744d172f30d712d"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2025-09-03 01:00:33.000000000","tz":330},"committer":{"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","date":"2025-10-23 03:53:54.000000000","tz":0},"subject":"[NetApp] Default node model to empty string","message":"[NetApp] Default node model to empty string\n\nIf a netapp node is down or in a transient state,\nit will fail to report model and this causes the\ndriver to fail init.\n\nDefault node model value to empty string if not\nprovided by netapp.\n\nCloses-Bug: #2121812\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/960562\n\nChange-Id: I544877f192375ab68a1bd8a547aa136c0137c8d5\nSigned-off-by: Hemanth Nakkina \u003chemanth.nakkina@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/711c97a1208ecca5c3bd96880677aa62d10cf34c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/711c97a1208ecca5c3bd96880677aa62d10cf34c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
