)]}'
{"id":"openstack%2Fcinder~959383","triplet_id":"openstack%2Fcinder~master~I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25","project":"openstack/cinder","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-04-07 17:12:24.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2026-03-05 20:58:51.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"38702":{"account":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"last_update":"2026-04-13 13:22:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25","subject":"Replace _add_to_threadpool with native threads","status":"ABANDONED","created":"2025-09-03 19:32:52.000000000","updated":"2026-04-13 13:22:21.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e69455fa89bf2a53a8600b4f8dfcd27775aba63a","_number":959383,"virtual_id_number":959383,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-27 00:00:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":-2,"date":"2026-04-07 17:12:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 19:37:34.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-03 20:54:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-03 22:22:25.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-03 22:34:08.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-09-03 22:44:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-04 06:26:15.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-09-04 12:04:55.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-11 14:14:03.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-07 18:37:00.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-09 22:22:38.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-10-10 11:57:03.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"CC"},{"updated":"2025-11-19 14:23:01.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2026-01-12 16:32:01.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-12 17:02:23.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-01-12 17:06:12.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-02-18 16:28:23.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-02-23 19:06:36.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-02-23 19:14:41.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-24 03:02:19.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-03-11 01:39:33.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-03-20 09:44:33.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-03-27 00:01:02.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"d4f3f84ecd4c2e6af410addafb71a7f65dce2a0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-09-03 19:32:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e0f9b7b93359d4f40e4ca42178ed76a01060d3f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-03 19:37:34.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"36e862b8ac311362ec2e6c1c6df3386171a80983","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 20:54:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cb35df76291465bbb0cca3ccd2f0295\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/33fb92db09e24f0aacbbe78177a3054a : FAILURE in 9m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/89c16d7a4faf4a968ca4b5f86966128c : FAILURE in 13m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4128beb83d364cf3a34441d08e91c790 : FAILURE in 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"311e4494a3d939d180fbf40de23f4291862259e3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-03 22:22:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f3b6fcf90c474ec5a4be20f3d3d69868 : FAILURE in 1h 12m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4fa162541d334a838f2bf31bf8dfd0b7 : FAILURE in 1h 05m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d4521ca194891d4e8f90720e527a6e5baff88fdc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-03 22:34:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 1h 24m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4416ea2660d1680d3ac557337fc3e20abf7e7e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 22:44:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85894782e8f84d02a982e41da169b3a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/994031317a594974b134450492056534 : FAILURE in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31478572f645444bba5bd4b7f0091e09 : FAILURE in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99bedf4fdcb14a7bb5812f4cf2c012d0 : FAILURE in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7783af36f8b44190aca1ba3c9d8138f3 : FAILURE in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ff5d23b2e85482b8ac384dbffde782d : FAILURE in 7m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/d51ac7e08bbf422c85cb3bf9f561a584 : SUCCESS in 49m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/63ec9554d4ee4564afb221566b8aacdf : SUCCESS in 1h 03m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/39467a61becd493cb7584730c09a533a : FAILURE in 1h 18m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/217d20e30db441d09687e19a9e2dbd45 : SUCCESS in 48m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6664d3b494994c5495f7ac027195ec77 : FAILURE in 5m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/64fc95483c1445d891b829f8a96f367e : FAILURE in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8de2daabf6b04abcbbe1ec42251a3502 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/02a117f069c64bafad91e7fa4dd6414c : SUCCESS in 6m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9b8b303e6b6c49059c05822fd87a9d96 : SUCCESS in 8m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0f9eee2823914f659e29ee78100558de : FAILURE in 38m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/65a3b41f1b4049dc9605fdb731f2a44d : SUCCESS in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9f11c1c916c84ebb8cda93c06cace1dc : TIMED_OUT in 3h 04m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/95c0bb71b1a7432ca900a9c3a2639ded : TIMED_OUT in 3h 03m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/527a4865e9824fb4877df117ecc5000b : FAILURE in 2h 25m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/66648bdc475d40b6acfa8fe64277776a : FAILURE in 12m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eeadd3c82099465dbe402eec73f486c3 : SUCCESS in 34m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f3cff66f73eb4e59b77d72b73e213ede : SUCCESS in 1h 01m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6a0d76c00e084ed8a5cc967bd728877d : FAILURE in 1h 47m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6fe9066d649d4d69847e37f2b392b959 : FAILURE in 34m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e059d5203cf64fa8995e8347392b3fdc : FAILURE in 1h 11m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/44b6579360e44ce2b0c4c00e973f0ac0 : FAILURE in 10m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ec108424312b46faac4aaba71e148169 : FAILURE in 59m 55s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9dcca4e1259740029462ed6191097036 : FAILURE in 1h 16m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2365c18fe2d049edb9938d1ffc909dfc : SUCCESS in 1h 01m 17s\n\nWarning:\n  Comments left for invalid file cinder/cmd/volume.py\n  Comments left for invalid file cinder/service.py\n  Comments left for invalid file cinder/volume/volume_utils.py","accounts_in_message":[],"_revision_number":1},{"id":"11a0dffdf9317c844522606cb11a1dca3fdf1de5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 06:26:15.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/433/","accounts_in_message":[],"_revision_number":1},{"id":"2da547835a6762512668a649f6afcf4ccec10e9d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 06:26:16.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959383/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959383/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"91ab4ace246b78a89b7897154ee9dfe1237583a8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-04 12:04:55.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1a81fe9f5deef71850fb12d83ed66d926cb4cc11","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-09-11 14:14:03.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/959383/1/20250911102256/NEC-ISCSI : FAILURE in 11h 34m 20s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/83/959383/1/20250911102256/NEC-FC : FAILURE in 12h 51m 04s","accounts_in_message":[],"_revision_number":1},{"id":"85e08dba2585ff1953509f76daf6f997a8f17f5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-07 16:38:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"442d9a24beb2dcee776d07095fda1e0c1a3aa920","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 17:15:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/079b4759707447588ca3503243503c96\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54f59ddc70a94a79a10c80f8c29121d9 : FAILURE in 9m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/edd4c16a22004eb18babfec94624c81f : FAILURE in 7m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6e64c99678c447afba16440b98da0f64 : FAILURE in 17m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a471d63fcfd7314da62d493c7040148214725898","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-07 18:37:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/959383 : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":2},{"id":"49de36cd78da2508ca2ce7b2f7dd12e95753234c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-07 19:08:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 1h 13m 15s","accounts_in_message":[],"_revision_number":2},{"id":"b596ec59417f29ceb5d2b09fe687cfe71006d725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 19:51:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4369e6dda97240d6903637d57ec5fb00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed1f62b41b24b47b1dc9142a93418d8 : FAILURE in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/660ab81ca9514b07883c480c68f7c964 : FAILURE in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/315197eee6bc440493e977c762744add : FAILURE in 5m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/231fde5d349c48f2855608a54191001c : FAILURE in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c6fc7de87464894aa1561ee24cc50c5 : FAILURE in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/c2f39cc900724e6d8b76846834154ba1 : SUCCESS in 1h 03m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ffcc2c64cf041f1945df5f2b8817228 : SUCCESS in 59m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d003c177c275471b8506d35c8c8ec0e6 : FAILURE in 1h 27m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52223f8502904481bd17f4c47c8f5d20 : SUCCESS in 1h 00m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0e7989c4031748e595cbd67e97283a09 : FAILURE in 6m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/868e921da96c414cb7102958ad57731e : FAILURE in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dfc1044eb26a48d0a29fb0476b083ce7 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a30ce5b40436477480a9593f5ebbef06 : SUCCESS in 5m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/46505a1bba8e41e3a7ff2b7d7d18c9d5 : SUCCESS in 8m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f1090a65c4884786a8a0b64c9b9214aa : FAILURE in 45m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/383344f0a83d45cbafffbd8169a6b7e4 : FAILURE in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e34dd1cc96d41afba6a887bbf9077b5 : TIMED_OUT in 3h 03m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/394d69d377ff4263a30954a851ffb8a5 : TIMED_OUT in 3h 04m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b14727153302477b8cc24cf43fb8680b : FAILURE in 1h 56m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d158521e5fef4306b0ad4576f17a6332 : FAILURE in 2h 16m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8e3d5807a543459fa4ef268259e2caf4 : SUCCESS in 33m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/faafb7f54c724a61bc5be7d0477f3023 : SUCCESS in 1h 10m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7a30f7207e8e4b0a9bf524457a15b768 : FAILURE in 1h 51m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cffeae40600044c5837d824264f1c86b : FAILURE in 46m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3feefb0c589a4fd1a7e6610e31d8f18f : FAILURE in 1h 18m 10s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/df2a4d2b1bf14f07846bc65b5815f28c : FAILURE in 1h 32m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3c654027ed14b8fb8d894ae49669024 : FAILURE in 1h 45m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ffac0335d094aeca79c0c895259c80f : SUCCESS in 59m 36s\n\nWarning:\n  Comments left for invalid file cinder/cmd/volume.py\n  Comments left for invalid file cinder/service.py\n  Comments left for invalid file cinder/volume/volume_utils.py","accounts_in_message":[],"_revision_number":2},{"id":"fd6af1c59e4c8b3097c352ebc28e5f10a2b74c27","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-07 20:14:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1d88bb8a222545dfa4a105f67f3cf863 : FAILURE in 1h 08m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a7b6ac0777654d5aa6220a86593431c6 : DISK_FULL in 1h 48m 42s","accounts_in_message":[],"_revision_number":2},{"id":"17144474a11fd66c0d74aebbf1512cef47e0d8de","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-07 20:46:58.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":2},{"id":"02dd53cc0fe607273edcece5a7964efa2b3555ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-09 14:20:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8620c6fe155e2b3fdd5820fa957809a1b0f83303","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-09 15:01:22.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":3},{"id":"287cb8b6bf29e735e75f5a3b4b2dbf40c3988fb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 15:55:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94032d4570574538829575efc48a4007\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8d0701b3494447ea89e8539398c63e63 : FAILURE in 9m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91c705dfa23d477cae163e92c83cee9c : FAILURE in 10m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f27641d354f449a7a1461618e88b798d : FAILURE in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2ddc14018d7c1a3f4dfc95b87480f72ad730b0dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:31:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c250f3fbad44e0ba00913fe21b08d3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20747239533c4f7481193190e84dcb75 : FAILURE in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88ec160b29504f73a9fa756274bced6a : FAILURE in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88279cf0946943f4b9fc88f58be2baa8 : FAILURE in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ae432486a424b82a9cb5401221c4371 : FAILURE in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a9598b6023d4bd6bb1572f4dca45a71 : FAILURE in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/9201b260233e4418a7f880110ffd9699 : SUCCESS in 39m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5dd8d040c74642f5ae59f22e654f9684 : SUCCESS in 40m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7739a64466bd4c05916be7020c778972 : FAILURE in 1h 29m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b60c6ca2d2b049c891689160c92084f8 : SUCCESS in 1h 06m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e5aa499efdeb47009c2fa3a7507004b7 : FAILURE in 5m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/77798f2b73ac4b3f93ad793fd97f7b1c : FAILURE in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ad3fadb971254eea85c8a5f2dca479a1 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5cad6a2eb8c043ccb0f33986d13b1c3c : SUCCESS in 5m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/010d3f97de3d4234862566d1dd286b8f : SUCCESS in 9m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ac97e7ba9edb4e568b73361aa0cc55f1 : FAILURE in 40m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/21dd94b406364c0895ee0cae3da19e8d : SUCCESS in 4m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6d2fc6aa7cf14b48b6ab69a3158fd2f0 : TIMED_OUT in 3h 03m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f640c74e139f409a9878e75f33c60a49 : TIMED_OUT in 3h 04m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/362f09683b2a4bcd97fb5b2c82d1f499 : FAILURE in 1h 57m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cf822f416181470b9e5826d880d5c5d7 : FAILURE in 2h 13m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d1e064522d5142dabad0c25e8a73f1bf : SUCCESS in 29m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8df31626b7f94ac39d14d66fa945ac4d : SUCCESS in 1h 17m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2c6891da649744b6a6291e1dea1d4c9d : FAILURE in 1h 36m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ae730b4a54394ed7ae500adf19f23432 : FAILURE in 43m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ec624d0353b54e52a818750456f941bc : FAILURE in 1h 23m 11s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/b6b8e57955444b5baf46e69c8c50fd8f : FAILURE in 1h 17m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa254480db164315ba69aa987239c4d3 : FAILURE in 1h 21m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9690a9bce121463b8cd1f6355d69629e : SUCCESS in 31m 39s\n\nWarning:\n  Comments left for invalid file cinder/cmd/volume.py\n  Comments left for invalid file cinder/service.py","accounts_in_message":[],"_revision_number":3},{"id":"f8d244bd9cebc5933bbb588127cd89b0f569f99e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-09 17:52:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/959383 : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":3},{"id":"8c677b76e96ec64dbdb01612c52a51e4a22f6c2d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-09 18:24:11.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c1c1c7ef032f40d28641a665776c56b3 : FAILURE in 1h 03m 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1f1b2bd0e39a49de8394f639584ad90b : DISK_FULL in 1h 12m 20s","accounts_in_message":[],"_revision_number":3},{"id":"6b1ae61a8cdd56749feae5b0f8c31aff57d96239","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-09 18:51:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 1h 16m 31s","accounts_in_message":[],"_revision_number":3},{"id":"ded795e8211ece32bab60608e5d9c9b90f11c366","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-09 22:22:38.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1970/.","accounts_in_message":[],"_revision_number":3},{"id":"7047803bd317e0045d75ed3a484d56ea93ab4546","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-09 22:31:56.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2084/.","accounts_in_message":[],"_revision_number":3},{"id":"83fcb1818efc91a1c9de1fc8f6db9864cdea15a8","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-10-10 11:57:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e3c58c9f11d94aa1e4a9ba32ee08bd902ee3fc7a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-10 13:26:57.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":3},{"id":"942842f239194c3f11d0c2f0d3d42b9c49945384","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-10 19:44:36.000000000","message":"Topic eventlet_dev_4 removed","accounts_in_message":[],"_revision_number":3},{"id":"03304743b01b998261837d4f3162266ffc57f919","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-10 19:44:50.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":3},{"id":"23c998ddb48c8dbc044613dd8d83a059030750c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-11-19 14:14:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1c45e1fab1d1cded0bf7092348347dd7ca5bdee5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-19 14:14:34.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2511/.","accounts_in_message":[],"_revision_number":4},{"id":"e595b1eb9dcab6e8541fd12c8369fcc35d157e5a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-19 14:23:01.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/7/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"e64be56ff4c77ef6aa65c36c2157a439c2d39cc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 14:35:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d786c54fd9c9431b8e991b2d865e15d1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc4c34f0f5e242a29897239cf7c709eb : FAILURE in 10m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fa5c2759a16475e80519f58a6336358 : FAILURE in 7m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9698f948e50640dfbc779316da9de54c : FAILURE in 18m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a4c9dd74ff19929cb4c80c64a07e286ab65e4db8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-19 15:26:48.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 1h 10m 40s","accounts_in_message":[],"_revision_number":4},{"id":"af4cd70c29065abab7a8fff0ce972589aab9a6ba","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-19 16:14:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e08e111e88804248b127c3fab88d828e : FAILURE in 1h 13m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/208fa63a7bce4ab5833f92dd35a31490 : FAILURE in 1h 19m 55s","accounts_in_message":[],"_revision_number":4},{"id":"66a32da3ed51b074543dcb3bac80d5d31e285346","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-19 16:37:42.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":4},{"id":"df1f2c5cf838386c21498c5ef4e536ee42c80397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 17:25:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bbd5c4ace8d472992daf01c0a1a6140\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13991c97db24405b424815d278a9b1e : FAILURE in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7208d7a2cfb46b388c23e957c9493c1 : FAILURE in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92d3437f2a634da6b840de633a4d0150 : FAILURE in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0085757202b6410e9bd465aee7abc5fd : FAILURE in 6m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/f439028920f74e129d8efc839527c77b : SUCCESS in 1h 03m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dda5057b7890446bb3cdba8c107d6924 : SUCCESS in 59m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/781f463b2710482788ceca683f7feb9d : FAILURE in 1h 16m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dda8d66cda849eeab29f6d2bfe3831d : SUCCESS in 48m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a20ca3307b91472692c0ba5239b60b7c : FAILURE in 5m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8091031d7ebc4e27a442dbe9f6f8be96 : FAILURE in 4m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03a6980b13174fa8973754b00260de0a : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e09c2b7f63bf49a29b38845ea329f320 : SUCCESS in 5m 18s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/675ee696c7a0407b99de32db85257104 : SUCCESS in 6m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/346fbf232f0f4f1e88c1c3ceee3411ea : FAILURE in 32m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5974e71aae20437da1b7ece2edcc890a : SUCCESS in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3cc4be956bbc4f8aa8afcc0455621f18 : TIMED_OUT in 3h 03m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/585e043d19814f8c96507b47bf1757e3 : TIMED_OUT in 3h 04m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/81c5842726764bc58ce438e1b86f2329 : FAILURE in 2h 22m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1847115dfa86478ea8a2d4f419fd2250 : FAILURE in 2h 32m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8fcf5e29b0c046398516e8b977f52a92 : SUCCESS in 36m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3753adb6dfbf43eebf01ab9b3fef97e3 : SUCCESS in 42m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/47d9fe8a554245bc80d2c71c492cf245 : FAILURE in 2h 04m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6756c59bd1a54d6ab1f1c399aed23f10 : FAILURE in 48m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6ac2ba186faa45cdbe1dd6e64cfcebb2 : FAILURE in 1h 46m 05s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f07ac9126900449cab86363820ffedfd : FAILURE in 2h 37m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6e65ff65fc5548fe98cf7a0551d64b99 : FAILURE in 1h 43m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4274d3ef52c945b6adf69b8860caf285 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":4},{"id":"3c8fb5fc26c169f6ed41be5d88118e8def44c5d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-12-09 18:19:06.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d0953b6f71000c5a0a88767132b66dd661e1fefd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-09 18:56:26.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/160/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"bf9103075c0639781ffcde60c72ca2e3fbc958be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 19:07:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62639df2d815495ba750cc85ab938012\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d80d22ab1bec4d1cbc900ab56a7d4793 : FAILURE in 42m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1a20ee395c7144b298f0ddd0e453f3f4 : FAILURE in 35m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a7eb1b78910a422fa0b92a3fded5abc8 : SUCCESS in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2acb75085fad518de30e6bc1d2ad333c949b01de","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-09 19:18:41.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":5},{"id":"35599f1300b2e9d7aaa8c39daa46c9ea5449de58","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-09 21:14:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2663/.","accounts_in_message":[],"_revision_number":5},{"id":"495e66dddeb3049fb5690334a9b958c750379ad5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-09 21:17:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b6062032c2d44ed88404f001509088f4 : SUCCESS in 52m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b6e6cf74924748628167a1a2c79bcacf : SUCCESS in 49m 51s","accounts_in_message":[],"_revision_number":5},{"id":"781c03f9aa6d434e5c7e4e33fe52a7d0c3e4305f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-09 21:20:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2810/.","accounts_in_message":[],"_revision_number":5},{"id":"660fb1fd1fc1c20ee075061761bc2098d5e6b40a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 21:35:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56f8fe233f6440538c9bd2c632c55e30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28418d31f144be9952bb6a5fa21eb83 : FAILURE in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62fef37db05c4f5588bf75a1ab1a7f76 : SUCCESS in 14m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f72c413689cd4c518bdb896cb3c6e80b : FAILURE in 24m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b46fd7118843f4aeef2000a8d5f70b : SUCCESS in 14m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/40317c8e63b24323b34f4c8c8325d665 : SUCCESS in 1h 08m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/91e01ea73eba4c8787a5e57738b1ed24 : SUCCESS in 1h 00m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b833386b68d40fe9e51395cec7ee7c1 : SUCCESS in 1h 25m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b89fdf89dbd42c88cb79e4db9e03bb5 : SUCCESS in 53m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a475cc51abad431389f81d9267f68ae4 : FAILURE in 32m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c9fed66013f04c37a4df0316550ef57c : FAILURE in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/efc7dade7435449581dd242e09372cbd : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c81c71d1270942f19cf730fed3f5b4bf : SUCCESS in 7m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f0f791f4909342bc93ae5c0189600df3 : SUCCESS in 8m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2556b7fdce57496eb5122d907c47a122 : FAILURE in 37m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a03a4cb0c20442e90c287a82f87185f : SUCCESS in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/448a128d3222447aa1f47bbb6f593c94 : SUCCESS in 1h 50m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/827e6e041a44466daf1ec82135883703 : TIMED_OUT in 3h 04m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3077deb0d0a24041a5cac820fe00a2f8 : FAILURE in 2h 53m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ae1f6259ac8747f0aa4640c6725e999a : FAILURE in 3h 02m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/53aa500e11084384a1899c0af0b05348 : SUCCESS in 34m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee49365241c3432693a953770fb88e7f : SUCCESS in 1h 09m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/088d843722f240f29cdea66103ec027f : SUCCESS in 1h 37m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e47f10aefc944050baf80213a0d66c1d : SUCCESS in 24m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/62b0bf36e90341b29b3537e61c7b207a : SUCCESS in 1h 51m 44s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bc592fa7d51b4dc2a4477b8f90065213 : SUCCESS in 2h 00m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6f69962db73c4ecd86facb22c1dc7e1b : SUCCESS in 1h 41m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29f3343155434ba1a12c2a742142f7e8 : SUCCESS in 32m 48s\n\nWarning:\n  Comments left for invalid file cinder/cmd/volume.py","accounts_in_message":[],"_revision_number":5},{"id":"137451f49282a58c03d19e592b32cf8c73871ee6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-09 22:41:03.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 1h 06m 26s","accounts_in_message":[],"_revision_number":5},{"id":"f8da5da93aa87da3640fe62df7df2cd150a385b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-01-12 15:21:25.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"dd5870aad5aa39a2f5e3fc5ae5311e04f2b49d5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 15:46:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/308b1ba9fe264f97a9444da98ca5afdb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1516a4378c35411daef25d39ec81bcaf : FAILURE in 17m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7322334db5e149889c67167d613db454 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/28ea0c7eb81040649a079d26d366b653 : SUCCESS in 14m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"65fddc71c1fedb94c26b6eae71cf7f2886a56e0f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-12 16:21:24.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":6},{"id":"e43a9ca5b5b76868223a55869c2dfaf1ea8be226","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-12 16:32:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/295323a0cc874b398d340b12f753d876 : FAILURE in 32m 48s","accounts_in_message":[],"_revision_number":6},{"id":"2890791dc6913292ed6d8d2242fb186ea0c8a916","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-12 17:02:23.000000000","message":"Patch Set 6:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/187/ : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"e82dce6dc75e6a1921c74934744395e193c1c635","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-12 17:06:12.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/15d1ed6b8b5a48cc9bea36489b2ac50b : FAILURE in 37m 03s","accounts_in_message":[],"_revision_number":6},{"id":"455a140cbb99280a03d2b34bb62459f2708fa730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 18:09:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f486b4c3712e4a4db045302e00d2a7e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d9d09d8674541aaa13c06dc6e570fbc : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd78663b164841cd8b9ba7a59efe44f8 : TIMED_OUT in 1h 17m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa2043da20b842c5be5b0c509997290c : SUCCESS in 14m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/701156a385854c6fb44e9ee14c30a659 : SUCCESS in 13m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/d5a098c6deb94dc2b2b187ce6f3f2b2b : SUCCESS in 28m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c67056cf03148a49e3f5b08cd712a48 : SUCCESS in 1h 01m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/db1fcb9d1700423693cb1c8a79625b5c : SUCCESS in 1h 22m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18679305f46d49a59f07b5af8d4d3e80 : SUCCESS in 52m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b7d15e4bd8da45ec94c384677ef7a2cb : FAILURE in 30m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d2a97f41720546ab965a6506af0dd56b : FAILURE in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7cd69c3a66a47ee9c999f134a552e90 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f6470fd060fc4af282eb3afcdd3b1d38 : SUCCESS in 5m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7ec0de789f544925a0440fbdbf1182b6 : SUCCESS in 7m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4c6a9c06889d403f838cdb5ade258a45 : FAILURE in 37m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5d59819a4c84d6e99967c668c13f163 : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/48cecb49eb3545038583c3113ac07539 : SUCCESS in 1h 59m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/23bd4eec9ae744aca2a4ee92a071fc72 : FAILURE in 2h 30m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eb3616829ab1439bb073bfddc67dec5c : FAILURE in 2h 41m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b3fdc2d1efc140809a9dd61fe399113a : FAILURE in 2h 33m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c7b874330f294485bbba46f3c2b9adfc : SUCCESS in 36m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/24674fe6fade4165823c0d13c90a573c : SUCCESS in 1h 18m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1f6e66c09a70481c9247eaf5da39df7f : SUCCESS in 1h 37m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/003a9b12a8914933b132beb50786b60d : SUCCESS in 42m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c3d797e563b14bb7bb84605c62ed5a94 : SUCCESS in 1h 58m 24s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/6e8eb5653a0a446b8d43377d6fdb425c : SUCCESS in 2h 34m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57d475c58afc40988fa25615e5240bc8 : SUCCESS in 51m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/600ed7c3fb904e05b8ea3ce6cf8fec11 : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":6},{"id":"0f5ef907a9bff58757c6f7f0c28d2b2a7b369c55","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-12 18:12:07.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 1h 19m 00s","accounts_in_message":[],"_revision_number":6},{"id":"2349056793e7c7f8cfdc384c747227a9d6d1125e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-14 18:28:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/add0640d889e4a9fb56c74b3051f95e9 : FAILURE in 1h 11m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/995400dfae5844a9850c3056dee26c7f : RETRY_LIMIT in 14m 56s","accounts_in_message":[],"_revision_number":6},{"id":"8fdb3e4455349178b871d7cb167943fcf5b3712e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-18 15:20:54.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a05689ac43f5cbb6aa72c302cd303c1740f003aa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-18 15:32:32.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4b39cf1bf354406a8e3ffe17d15b5777 : RETRY_LIMIT in 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3037fce598be4522b6349815c1119833 : RETRY_LIMIT in 32s","accounts_in_message":[],"_revision_number":7},{"id":"2584c154bf3e23ec2a4e853f5f83231fb39b2c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 16:26:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebf0faa3a98145a9817150a8a163de86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2f1ad94da5b48cea0aab3b65d55f26f : FAILURE in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a9632b959d94b6e9809645dad714b0c : FAILURE in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0a08d9e5c2849e09914dca3badf9994 : FAILURE in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff47a69a744b47ef82a0d7ca53c48f8c : FAILURE in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/8522d6c307df4189b47514e35213112f : FAILURE in 55m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7bd08ae0b93f4be69acf4fa4d9c430ef : FAILURE in 30m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a408dfc472aa43ebaffebe9e44270888 : FAILURE in 27m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e569d48abd74d7f9fae8297dff05e18 : POST_FAILURE in 14m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ab557a2c585f4eb2b58809e2fadbd042 : FAILURE in 5m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eb779304edec4a5782822745014ba3aa : FAILURE in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86a409f92b0f41ad8b78ab5994a8a5ec : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f0419cff4914f3cb99f15389d323b3f : SUCCESS in 5m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5a385cac608443afb2c6cf6e7088e39a : SUCCESS in 4m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bf84c6a2e0744a2b81656cf1fee67266 : FAILURE in 27m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3608e000652a4ff8a20d456c35b49ae7 : FAILURE in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8588613882ec4335a830a09bec81c892 : FAILURE in 17m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/88477c2bbb52449a9b68297e6e26a904 : FAILURE in 22m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02e312c2b3474ef0b862b10b60bc9a76 : FAILURE in 31m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2967e75adb6d4097a6464b2d5a9827b5 : FAILURE in 42m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/913c7d7ff71542e3a6d632357e3bb057 : FAILURE in 20m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6e985d3471ef4841a786936a52c2d5df : SUCCESS in 41m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d1eda23901ac4e2b919ec8f0fad45f92 : FAILURE in 24m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5460c90a98fe4050b500fe74453209d1 : FAILURE in 28m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0d7b452d19a34598bb7d070e3bb7fa0a : FAILURE in 26m 34s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3f242d5ee78b48ca8735182a9627a6be : FAILURE in 37m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/290035575e9d4cf9886ed9638a2c895c : FAILURE in 16m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0417dfdef2144d5ba93618d99e823c3e : FAILURE in 13m 38s\n\nWarning:\n  Comments left for invalid file cinder/cmd/volume.py\n  Comments left for invalid file cinder/service.py","accounts_in_message":[],"_revision_number":7},{"id":"ba0d21ad18b4245df0ef0cd42caea9dd1e968658","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-18 16:28:23.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver build/c2abab8f5e35407a8e36052d4db71ddf : FAILURE in 33m 40s","accounts_in_message":[],"_revision_number":7},{"id":"ed4cae05e54deab108fc7901bcd831603cc04666","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-18 16:43:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 39m 26s","accounts_in_message":[],"_revision_number":7},{"id":"f3603cbf4b8bb49e50fe5358045cee1d3267ca30","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-18 16:44:26.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/6280fd6d68f44703b8df7d20fade8313 : FAILURE in 39m 41s","accounts_in_message":[],"_revision_number":7},{"id":"001fdad82c8d78cb0f88d6dc5eeb866000a0d8e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-18 18:14:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"1588892652d0294ac4d60066d0495b5ba39c197a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-18 18:25:10.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/faa75f8a956f49fea1ff358be0a9dd8f : RETRY_LIMIT in 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4fa5a040baef4fc493a5c6cb42e27f17 : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":8},{"id":"112ad1b76a774f9d7b7d432e88af179cf2cf5b81","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-18 18:42:49.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":8},{"id":"46cbed2094b755dd79106f240fa5fddb47c98956","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-18 20:06:41.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver build/37e72daa389a418c8ef4d0fb4d2725ce : FAILURE in 31m 20s","accounts_in_message":[],"_revision_number":8},{"id":"30c292fe857d6a6f44a66179213bd365a310bd80","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-18 21:25:38.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/6b62a94c67274483afd66748cdc19be5 : FAILURE in 43m 02s","accounts_in_message":[],"_revision_number":8},{"id":"817f9d5dec4d13708ff20e85282dcafc002eb9fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:32:08.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/307045a8301841adb90d8b6df5d4d602\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f293983043f34566a0f48fbe24617c57 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7411da9363f4ed2b7d2b7e4f740e659 : SUCCESS in 8m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfa1a798a42546bf9cf718cd836b9946 : FAILURE in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff657ef7ac248e280bc3859bd9d04a1 : SUCCESS in 15m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/65c3f441267c4a318a988fd34ab83b38 : SUCCESS in 58m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/32b5e9b41e6a4a5dbe83a05044e850d7 : SUCCESS in 1h 01m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0eb46739344b4dc4ae62277363db157f : SUCCESS in 1h 23m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3037bc46fe5475a90ed1bf81169c27e : SUCCESS in 1h 03m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e49543297b2e4db2a14aa6032a6ac13c : FAILURE in 30m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3b8cc61fe44744bbb15df863d898bd7e : FAILURE in 3m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8480ca8ea72e44b4bc04fc53b1452f55 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/84f668ef4a1f4b448fddc22fbb9f327c : SUCCESS in 5m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/458ac3967c8147ff86a4b3c6a1893eb4 : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/75faeeafd80d4901950a15e091ba7315 : FAILURE in 42m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/524a3d0354064765bf64bff8edb38c36 : FAILURE in 3m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a580814db0f4f3abd3bd7dcf77305c2 : SUCCESS in 1h 57m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bf88c69272d845c0874fa1badb93f890 : FAILURE in 2h 14m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9932db28aa374b18a86d6fec1be8de2b : FAILURE in 2h 45m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/657ab8695fa943b2a3c5ba3cf24b6a6b : FAILURE in 2h 52m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a53bc7a90bc34edbad6ba9f0d26a8977 : SUCCESS in 20m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4fdbc60618af4b46940b07d20d3d5d97 : SUCCESS in 41m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d8f3c23bfee64f328ea5e2ea5f8414ac : SUCCESS in 1h 32m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dbef0af7bc804f9cbb6d787e17f418b7 : SUCCESS in 41m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e9ee91581b954b0e8d60a9d11443bdbf : TIMED_OUT in 3h 09m 44s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bc433ebdaaea4b7cb5b97640357b2c7b : FAILURE in 1h 30m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/591da78312ce4240ae6f134afebe8d5a : SUCCESS in 1h 44m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74407c599bf44403b7132de8742dcbb8 : SUCCESS in 32m 16s","accounts_in_message":[],"_revision_number":8},{"id":"c8fea28523332a3ed98dfeeca74295b5f9950494","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-02-18 21:50:00.000000000","message":"Patch Set 8:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/738/ : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"f51f26cdea96f003e980705880989bf7b7fe5712","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-02-18 21:50:14.000000000","message":"Patch Set 7:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/725/ : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"24eed763800c3ed2c15cc6423550a58eff7726f4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-18 22:06:50.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 40m 04s","accounts_in_message":[],"_revision_number":8},{"id":"e979867f028aa6e58b02861af9bba9e4d2d4612e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-18 22:11:11.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/202af7ebf0a04a00934a75197be2dfef : FAILURE in 40m 36s","accounts_in_message":[],"_revision_number":8},{"id":"6c17b8fcfad82f22ed1ce334bccf7fcfa9541a7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 23:18:29.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aefda6339bd041319ce24ba88fc8af2d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ff0d37a7e3384b45851d5b25c941d63f : SUCCESS in 16m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d540ad56383a40099b327eb2ca18a2df : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1d1a1789b72640c8ad31341b106295b8 : RETRY_LIMIT in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fc58faf7971136688f553d6a540bfd54970370c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-23 17:46:41.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"d358720b91caa37dc16005bddc5462ddc1c541d0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-23 18:18:21.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/928382ba76f242b584bd069a1590e91c : RETRY_LIMIT in 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c9dde3d9d2bd41838ac65d5011d32049 : RETRY_LIMIT in 35s","accounts_in_message":[],"_revision_number":9},{"id":"fca7165c3e6d73889b0859c82cf5e4e4efd7098c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 18:22:39.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85cdd5e80def45a8aed832249d05449c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d3a10fed3b734c2cb697d24ce5f53b27 : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ce4db83742a446dbc9bd838ac324932 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3434fbb7892494eb177fe077dddd26f : RETRY_LIMIT in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a5664c67b566ae26da1b794e95b8a87eabd296db","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-23 18:27:44.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":9},{"id":"9331b4e138d614cb6f944f42c51b53f5856dae35","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-23 18:36:21.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/26057d92cb04463d925d8423e740bde0 : FAILURE in 43m 50s","accounts_in_message":[],"_revision_number":9},{"id":"444ac0324990f1ba4b6f824cf193b488b4f70c53","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-23 18:58:04.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver build/5baeacaf1eab4f8c81c4216744722875 : FAILURE in 32m 16s","accounts_in_message":[],"_revision_number":9},{"id":"816ea82882a39c43411db223b21c2679ec0ca0b2","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-23 19:06:36.000000000","message":"Patch Set 9:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/959383 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"b42d4e663e8c97bb70a580c09ebe05117e8bec4f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-23 19:12:53.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 39m 15s","accounts_in_message":[],"_revision_number":9},{"id":"0e0597d763c697d2b38a1013c54c5d372f8d24e4","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-23 19:14:41.000000000","message":"Patch Set 9:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959383 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"5f8b21e8eae8d83403f44d555381ea5d5010f898","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-23 19:19:24.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/47bb0c99286740cab893f6275ebdb1cf : FAILURE in 38m 40s","accounts_in_message":[],"_revision_number":9},{"id":"541492b9895b88140b4ddb6a650ad42f5721bb09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 20:13:42.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1c9e36381f44677ae2c592150419acc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0a7036f3374b84939a41e871056908 : SUCCESS in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02df345cbb74449d8dbcc3fc878a7396 : SUCCESS in 9m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7afad843f81474483bff08a7945bb09 : SUCCESS in 12m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/00d0abc3af59455a883b6375b7d51f2f : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f88a29a438049d2813c7aeee08ae520 : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/df383e995d5640d1bf389705b497ebff : SUCCESS in 51m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b6d71cac453c4c7b85a52ff684f41725 : SUCCESS in 40m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ebc02b4951f4033a87703c783d81cf4 : SUCCESS in 1h 20m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8498fd44c97f47aea707d072c379c1da : FAILURE in 1h 12m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0d59051d902b4f88ba0a9cdb57b60564 : SUCCESS in 22m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/92886f0ca9754c0cb2349ee2164da9ca : FAILURE in 2m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e4499815b2248a6a1d92f824f0d0477 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8eeb8d64ef74c4dbb3daa8a98190d2b : SUCCESS in 6m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f3f9c18fe6d849c289e9c0e7a898650a : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f5f517c63b1e48bd9e94bfcddcee9865 : FAILURE in 32m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8fd751268448482188b30156120898c4 : FAILURE in 3m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/97bdef6a196d466cbb9f466cce1d703e : SUCCESS in 1h 41m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fbd11a07197a48a0811dadad6aa9d5ff : FAILURE in 2h 15m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e78936788e4546b9b85b1ea2ad7082f2 : FAILURE in 2h 01m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c209d73a6e0f45e2af327db28a91c95f : FAILURE in 8m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4c820ad740c442a1bee5419e3b767b5e : SUCCESS in 29m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/08af22eb1d9845f7ba4814363b482796 : SUCCESS in 36m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e7772468086f4973b31af7520c3ce4f3 : SUCCESS in 1h 27m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/99bb9e37d7ca438ea8460e6a3784dc1e : SUCCESS in 28m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b1b832259b344f54a03aa926132f9cdc : SUCCESS in 1h 03m 42s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9cee6d4ed757477782983762a7bc0bb0 : FAILURE in 11m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/168996833d8d46a59e36b60246842ac4 : SUCCESS in 55m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7425cab554748f8b754fe64198d585a : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":9},{"id":"b6d839bee37eab524df614ba55a68bafe1ab167a","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-24 03:02:19.000000000","message":"Patch Set 9:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_824/index.html : SUCCESS in 93m","accounts_in_message":[],"_revision_number":9},{"id":"392ca75719561522438af11f2b8cde2b7268fbfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-27 20:02:52.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"c48bbf951bf01433111ff0be165cfefad7eb5c5b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 20:03:20.000000000","message":"Patch Set 10:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3094/.","accounts_in_message":[],"_revision_number":10},{"id":"5407d55fefae087913c637c760064e10dce7365a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 20:22:32.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3258/.","accounts_in_message":[],"_revision_number":10},{"id":"24cee3093dfee4b4cf542a1e4ceda515a1f8ebe0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-27 20:44:52.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":10},{"id":"38a99b020120bf4511d9805ac0284041fc10f84c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-27 20:45:44.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a8f093bdcf85458cb91f59a83b501870 : RETRY_LIMIT in 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a9c7bb4503fd41c2a9325a968149458f : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":10},{"id":"666c83f51a13785dc4821e9ef075739e45a178c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 21:27:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/150d40741a56405391746a56c654dc4d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af58687b51ce415c967072848ba6a0a7 : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/676e93cd5e9a4db9b23eba868013971d : RETRY_LIMIT in 3m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/46729b6d46c94178a77f79a7ab28726d : SUCCESS in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f56fb3450ce05ec4c89a7301d339f3a7bde4e4f5","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-27 21:41:29.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/3964f3139fc245afbaecdee95c1fe44d : FAILURE in 43m 03s","accounts_in_message":[],"_revision_number":10},{"id":"4ee3fe0e779e0daaf969f417da678046cbd7ffb1","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-27 21:52:21.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver build/dd22252369884bf9a5129d35ad24a86d : FAILURE in 31m 01s","accounts_in_message":[],"_revision_number":10},{"id":"094f75799c53199db3263519b6266761aad2ef92","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-27 22:04:43.000000000","message":"Patch Set 10:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/959383 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"e7f07fb55109b2a5e3331399c30e6744ae9b287f","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-27 22:13:45.000000000","message":"Patch Set 10:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959383 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"ccddd3767e9749e8b0080d6866a80e5db61cabc2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-27 22:16:59.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 38m 04s","accounts_in_message":[],"_revision_number":10},{"id":"b442b4a12da8c810f14753bf04def7b77f4c2bcf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-27 22:29:45.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/362f21af89f349f39f41221e9f031c04 : FAILURE in 40m 34s","accounts_in_message":[],"_revision_number":10},{"id":"bfffc9e13a7592ead43b95d3a3b2ea81d3cb9a22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 23:04:39.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4d1a7215b754c9b98942adef6c0a29f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0ef32e022ba4a6e90a7047b8aebe204 : SUCCESS in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b4bcd1cf2be4b9294fa1e3451b09181 : FAILURE in 12m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3d1f5f2d34e4b25bb19ec64d1ffd7ba : SUCCESS in 10m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae4df5ee41084ce48554029a9a4dd214 : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656949ed8a5642a4bc11d8920775472e : SUCCESS in 15m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/2609378d71274769b48f5115f5b6412e : SUCCESS in 1h 02m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/265fd3a345644a8a9aa00c86dd93bbdf : SUCCESS in 46m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/caa479df2a6a4a249640a920fdcc9127 : SUCCESS in 2h 06m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7763dc26b5c84b14acf266359eb7d546 : SUCCESS in 57m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/32f048591f2f4810963e6cd6185844d5 : TIMED_OUT in 41m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/588b391b9fa2488c85d1116c091b469a : FAILURE in 6m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c427c86a38e64769b537c5c5f33427af : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e7376e754e7342a38b19808e21f167de : SUCCESS in 7m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/28d54c5bbada432d8e8ce9853731f891 : SUCCESS in 5m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/20473304d5c24faeb74c8f20b9e88532 : FAILURE in 32m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/148b9d30e4044c70988edf7bd42d4b43 : FAILURE in 5m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/685e8602e66943a3b180c89256b6a834 : SUCCESS in 1h 11m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fed0f908d6dc439c8b771d439b6595a8 : FAILURE in 2h 49m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5e0d77800b79406287bf508b388e67db : FAILURE in 2h 16m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a979aa5e6f5d45edb1316b2b2439f61d : FAILURE in 22m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2f0232d40c614e1a89cfc3b3605176d4 : SUCCESS in 46m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5c867dac69604e7eaefbcff40bf79b66 : SUCCESS in 59m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d3903417f8b14dc6a49c593f72ac4906 : SUCCESS in 1h 26m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f5c5f856eb3d4814819a20464e481dd8 : SUCCESS in 44m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b599790065184551b13746e77b6b52d7 : SUCCESS in 2h 04m 52s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/eac618df947c4727958fbc88492070b9 : FAILURE in 9m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/05d19d47fb6b4d879c9f1ba50fa31564 : SUCCESS in 2h 03m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0783eaae043a4e7187d5a7e0db5992b3 : SUCCESS in 1h 17m 35s","accounts_in_message":[],"_revision_number":10},{"id":"dc75d6a1dfd77aaf15ca0a0cc1c66d9692d80f6e","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-28 14:32:22.000000000","message":"Patch Set 10:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_889/index.html : SUCCESS in 90m","accounts_in_message":[],"_revision_number":10},{"id":"3c5cb4145b374e5a276a2be0d3f2ab1fca00d2cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-05 20:41:41.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"ab7f420de330e0819d530a7cb5e015e4fbd6b142","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-05 20:58:51.000000000","message":"Patch Set 11:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/959383 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"675cd56f19806422e8f950fd61021e2da03e38c7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-05 21:09:09.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/43b0a46c205544d9b146552451151bcd : RETRY_LIMIT in 42s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4215ef8546724112b0c58508ea512eb1 : RETRY_LIMIT in 38s","accounts_in_message":[],"_revision_number":11},{"id":"26b18d07c5b66a5882c957f015e3f957dc895ea0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-05 21:18:04.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":11},{"id":"885c9ad0e61479aab7267bf916776252d62406de","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-05 21:24:57.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver build/06548a98c09443e594c3cc7a51f0c2fb : FAILURE in 9m 39s","accounts_in_message":[],"_revision_number":11},{"id":"1c8ad9e19e060c0d608928dddb1bdac870e1aab7","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-05 21:30:59.000000000","message":"Patch Set 11:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959383 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"dacfd92b543424dad57e91f1a8dec6789c4251fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 21:40:42.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/709addeb26f4441b88fdcc7123e45b04\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/07832fb2902043cca96e79ec4ccbf81c : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aaeadf3d58db4994966707977141db2b : RETRY_LIMIT in 4m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ce454df4b4ca444b9e40ee07e1846103 : FAILURE in 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a65c2f75a166cebface92741857795ace7d1ea2e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-05 22:08:07.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/eedbb437e7f344148a3dba380b33af22 : FAILURE in 18m 17s","accounts_in_message":[],"_revision_number":11},{"id":"f743b40e2fa1c49ba7a1ab72b2dd928911bb9eb8","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-05 22:25:54.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/7e12d68468e6447381a9813a0e03917e : FAILURE in 11m 38s","accounts_in_message":[],"_revision_number":11},{"id":"ac11fddb56ddd92f4e493d5fd3632595c2ad47c2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-05 23:15:24.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 38m 29s","accounts_in_message":[],"_revision_number":11},{"id":"9f9c81666aa852622b2c34bf680716c0cb3f717a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 00:03:15.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca4144b3c8ab47afb010aa0a056f3153\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/847a14a191cb410688c94ee8e9abdab4 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ae04bfa6b8b4d0481f40bdbb94d3252 : SUCCESS in 12m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/785310c5fe3a48699f1b49dbe515bd1b : SUCCESS in 8m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e73ab3a692f748c4873d88ac56002657 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c7a4c373ee44beac00449e54b0f922 : SUCCESS in 14m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/2a49fc3f275c44fea29e175ed4dd2662 : SUCCESS in 56m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d7699a58a70545d3ba271c7932dd03af : SUCCESS in 1h 08m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e9243d6787a45d89a1cc31f293a6946 : SUCCESS in 1h 40m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db1c365f996e409791b0d47996dd4357 : SUCCESS in 58m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d12643377aa479db73be230085d5ca7 : TIMED_OUT in 41m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/19c0783c17444245ba85b9a502b21a83 : FAILURE in 3m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/496a3a7bfe36478894cb73188bd5a23f : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fd212e97adf54152a67898566fe60761 : SUCCESS in 5m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bc028679b2084bf6bbce1313f2ea9896 : SUCCESS in 4m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/77dedfad16604a1faf510735a1f27467 : FAILURE in 38m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d62afc6561124d79829c0f470d7158a1 : FAILURE in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4d4357e8ea704ab3be14c5b1b47601e0 : SUCCESS in 1h 02m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5ddbef762e9a4bf68fc474386075b608 : FAILURE in 1h 27m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e7c47d73740c4786a0efe357c46830d6 : FAILURE in 2h 13m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4c08acbd206742a1971863cd0adc8c9f : FAILURE in 2h 51m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a96b7200eea84372b7fac1f4c899bfce : SUCCESS in 34m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c1d6859ad43e440bafc0e12a5b963683 : SUCCESS in 1h 11m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4bc6da9e220d4084b5c4c1bc8ee86040 : FAILURE in 1h 49m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1dfb909d8320475cbe8c3eed917a5abc : SUCCESS in 36m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/86b43e2012314a3ea649edadbb5c8f9c : TIMED_OUT in 3h 09m 35s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/74f849356b1747d486248863b8ab44cc : SUCCESS in 2h 11m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/03479b997cba4f6e86ce5660db81a885 : SUCCESS in 1h 38m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6763a7939b154ba4b59c7ac6c569bc1e : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":11},{"id":"a4ed1e4c2a51d0328062eacdd49e55a0b35ce00e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-06 01:10:51.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3176/.","accounts_in_message":[],"_revision_number":11},{"id":"a6e7ef89c36c25bfeec4c238124d70f5a6738b51","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-06 03:49:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3349/.","accounts_in_message":[],"_revision_number":11},{"id":"110f10f8649c8063c4fc1f16f0bd38214878e588","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-06 18:47:40.000000000","message":"Patch Set 11:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_974/index.html : SUCCESS in 87m","accounts_in_message":[],"_revision_number":11},{"id":"fc422eb3dd33cf3b3b3ca21276233cfa76f7a817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-10 17:42:03.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"1cacd1241262dd890e89a2c91eaa455673a4c3cc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-10 18:01:52.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d6b01d0105804039914864ccebc2fb28 : RETRY_LIMIT in 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/84d0efe6cde24e068a9f82b9941c6f8b : RETRY_LIMIT in 37s","accounts_in_message":[],"_revision_number":12},{"id":"5e95726674c6bf9a3d727f42bfa19daf44ef078a","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-10 18:11:04.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/56920ac6d79b468c9629a4f7d0d25ab7 : FAILURE in 12m 25s","accounts_in_message":[],"_revision_number":12},{"id":"930cda024b2f1cd4aab1db473d8abce4b23813c3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-10 18:20:43.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":12},{"id":"dda14a0c3bca21d5c150170a673f53040c61534d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 19:56:02.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/262ef5eee81f4d34ab00113e78cbc68a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/121a00adaf9a485e8476365fa047ac3e : SUCCESS in 18m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c4ea2df39f77417bb047db16639b876a : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b602697564de4f39be1474c73cee4b40 : SUCCESS in 20m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"08ada7a6d26c2aeb5ba7110b701a24e056af5724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 20:59:05.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/962e0e245d7e4164901556eb88f92391\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a0a06c24f384c80bf7d3ffc4bc36421 : SUCCESS in 7m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8836242332843d08b665f3ad68968da : SUCCESS in 11m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76c72329c69e4822af7502cd53a7c24e : FAILURE in 13m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c7cff68c15e54f718a4e08419e5516cb : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c19785eb7f4a2f86e197462441f79c : SUCCESS in 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/297fbe9358344018a9dff304e42b123e : SUCCESS in 1h 03m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cb1ea158f9d24b8ba23a2194f53b3913 : SUCCESS in 1h 04m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6db2600e74024b799339c82cca37eb93 : SUCCESS in 1h 39m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83cd896dde20471ab301dd5346e299f0 : SUCCESS in 1h 08m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/01ad29da3f774e999928f3ed6d458bec : TIMED_OUT in 41m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3b833bdc71414d2d9921a761a5694113 : FAILURE in 5m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1f97e07f08d14f8687e9f7926ed15d4d : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/deb3cecef7954a2f973bb08bf9bb7c89 : SUCCESS in 6m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e90ee654ec594b05b3cdf3cafe5a5082 : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8e0a93d6ded341f89835772152487033 : FAILURE in 43m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/da7a2ff9c43b44d5b78fcceeb2a42ab3 : FAILURE in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c710c37aab6f4449ba1d77479e0bae74 : SUCCESS in 1h 46m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/66a24387b0c04495922e2c256e9a0ac3 : FAILURE in 2h 26m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dd2c5f850a3144b8879e0a39d18e95bd : TIMED_OUT in 3h 07m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/37a6bbd0d5a3476c81a883f012e805b8 : FAILURE in 2h 16m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f3b507b6150b4137a00a111280e63b97 : SUCCESS in 36m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3b1c2e7e1f65488b89bd37453b16a656 : SUCCESS in 1h 26m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b5bb818d17b64dd8ac5bb7d96bb83de8 : SUCCESS in 1h 32m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/af8d8e2286ea480b839dd44ec8d2b167 : SUCCESS in 47m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dedc6e068088466a8e4b34562aa0273f : SUCCESS in 1h 56m 33s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/7f9d7eb256ef439f89eb1d8b0c2e3d0c : SUCCESS in 1h 56m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/500fe36a3f3644c99e83882f85a38df8 : SUCCESS in 1h 50m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9e1d12471254f73bd0359a726ca940f : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":12},{"id":"55a9caa73b2c28ed5cce65db15c8b8c30868a56a","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-10 21:16:38.000000000","message":"Patch Set 12:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/959383 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"926248c7db0817ee1844af9aaed347e3a72092b5","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-10 21:55:26.000000000","message":"Patch Set 12:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959383 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"9a05dfa1f6753ff5a62285fb685dffff61d47b08","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-10 22:03:09.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 37m 34s","accounts_in_message":[],"_revision_number":12},{"id":"19adae33d9c858697aec552f3e9872746ac6ae96","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-10 22:29:23.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/7c3d2e9fbe21438da94b88a65c442665 : FAILURE in 39m 36s","accounts_in_message":[],"_revision_number":12},{"id":"8d6e278d490da8a428e6bb00ad59c92ea2d82683","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-10 22:51:15.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver build/96f2d2c660e54cd091feea291fa481fa : TIMED_OUT in 3h 00m 48s","accounts_in_message":[],"_revision_number":12},{"id":"67f026aac3b505f35b7946dfd2e7ca7e2337f1da","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-11 01:39:33.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-959383-12/FC/ : FAILURE in 6h 54m 51s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-959383-12/iSCSI/ : FAILURE in 7h 29m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"dc88ee098ab63e4af5423d0d4bfe231b57b99233","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 04:06:14.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3255/.","accounts_in_message":[],"_revision_number":12},{"id":"4ebd2841a90f4e23d85d8d6e7b2f46114bef3f87","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 04:39:03.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3428/.","accounts_in_message":[],"_revision_number":12},{"id":"a18cd9a2a43d547e5d2db26035a9aaa2e7af0d0a","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-11 15:43:21.000000000","message":"Patch Set 12:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1054/index.html : SUCCESS in 94m","accounts_in_message":[],"_revision_number":12},{"id":"c60416190d999587e260f86c2aa11fa1c3c701c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-19 20:43:51.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"cb07f0a5e3786b3fddab29f3e08250a9bfdef668","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-19 20:47:38.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/580d96312680440fb9807c7e9b31b05d : NODE_FAILURE Node request 200-0000658562 failed in 0s","accounts_in_message":[],"_revision_number":13},{"id":"79960541cd557c67bf9920e28a9fccab06ea5dfd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-19 21:10:33.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/abc401f9f84e4637af4df8cad914bf11 : RETRY_LIMIT in 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/033d1d1569fd47479674bf1cff449486 : RETRY_LIMIT in 34s","accounts_in_message":[],"_revision_number":13},{"id":"ced6f810ba30db26f77e3ef850dfb15f88ad51b1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-19 21:25:43.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":13},{"id":"5e04e23dfdca1d4a352ee8435c7b4baf2a33ac8b","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-19 21:53:55.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver build/e3ed877a39d045e08ce4035eb2e00683 : FAILURE in 32m 11s","accounts_in_message":[],"_revision_number":13},{"id":"20773a62f6a0509b5d7b4eff32908a2acb34980f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-19 22:07:21.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 38m 48s","accounts_in_message":[],"_revision_number":13},{"id":"5eac17a2a69cff0097971b6c155dbf42bca09edc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-19 22:14:54.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/442317c931294eb4b800749b201b363d : FAILURE in 39m 00s","accounts_in_message":[],"_revision_number":13},{"id":"63e3dd6ca87c5c069168e14bdcdec1dc9280eb50","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-19 22:15:06.000000000","message":"Patch Set 13:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959383 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"54f2055d21fa460d3a323941e71756218675124e","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-19 22:28:45.000000000","message":"Patch Set 13:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1183/index.html : SUCCESS in 91m","accounts_in_message":[],"_revision_number":13},{"id":"e895dd58a2fc49f3ef7d63cfac98f16dc0419d66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 23:03:41.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7838bc4de0f4516af8706cd4a77902a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/df7d3dabf4d34f2b9ac8606f5d0f64fc : SUCCESS in 20m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/496eac50168e4f97b31d845f3ed5b7be : SUCCESS in 11m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/912569adb2644cff97277dc1975a1de9 : SUCCESS in 23m 57s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8c31a6ff03eeb0f9f96530a7ba6b91a4ff359197","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-19 23:34:34.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3534/.","accounts_in_message":[],"_revision_number":13},{"id":"16262960989d5ff6bcc7f7114dc576e00932b42c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-19 23:34:51.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3361/.","accounts_in_message":[],"_revision_number":13},{"id":"dd95ca22e365db118c44dd147c8c6e82dc242a4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 02:58:38.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/796c440ae8cf4055a5263af843d94eec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8963912dc47a49eb830ae7ba031d2d8d : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7a791cbf537485984bbbaac8c71ee12 : SUCCESS in 10m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8abf0e125714be99cfdfc1349edb3aa : FAILURE in 14m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8da4c60553e54107b8b10fd8c34a911f : SUCCESS in 15m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6375fae0c5c4e1893800aa3a86251f2 : SUCCESS in 13m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/5c4be27c77cd480f9631ddfe71b37465 : SUCCESS in 57m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec2af0ccfa90422795dab7726dc9c558 : SUCCESS in 58m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/addd2f0ecb0448b0b60db42d267855c8 : SUCCESS in 1h 42m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/855d6b2c2b4547659ca78cf8e0e2f119 : SUCCESS in 1h 02m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6546b2a406fb426fb6486e1e9cc0adb2 : TIMED_OUT in 54m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1a14b7f88efa452ea5271a35b156d1d8 : FAILURE in 5m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93b4ec840461481a8376eb2e57c818f7 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0dc6a35d0d34621acddceea6eabd63a : SUCCESS in 6m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/caff385b442c402491ea88d849138884 : SUCCESS in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/af65c18d01e544eeb818d1dbda8a47eb : FAILURE in 39m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/146342adb82c447f9262f26106ef2ade : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4709e4f11ea04ed684d3554e0abb5be4 : SUCCESS in 1h 34m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/691b8107f3bb41d3b993433adecb8f12 : SUCCESS in 2h 17m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8f9c63e801e54e489f5402568164ee00 : FAILURE in 2h 27m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2cfc7ecdb4564e7794095a3121587ae6 : FAILURE in 2h 35m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dffd9e7b83294e389789c7d954d218bf : SUCCESS in 39m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5b128e71cc184e7ea6171b21670c347b : SUCCESS in 1h 12m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cb9a6a3a53d74f628d2a3848b30aa7a3 : SUCCESS in 1h 35m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/359b6bb2d1964e29b30a092adfef04da : SUCCESS in 44m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7fb8442d2f4e4ba997bc0967d4ce7dca : SUCCESS in 2h 13m 12s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/4ba958d558ae4807a4128116da983ef5 : SUCCESS in 1h 05m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9efb8be083e4dafbacad3770851214c : SUCCESS in 1h 35m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cae0cdab4fe40a29e8ae78f7bdca909 : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":13},{"id":"27ac7189e74a53880754893f289cbb494e11cf83","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-20 03:15:01.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-959383-13/FC/ : FAILURE in 3h 30m 0s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-959383-13/iSCSI/ : FAILURE in 6h 30m 46s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"71e7281fd5fb77c8d2cae354b996bcfc0abec46b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-03-20 09:44:33.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-83-959383-13-fc : FAILURE in 1h 4m 4s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-83-959383-13-iscsi : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":13},{"id":"6f77ab1293b534377d49fa8c767b13609de83fb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-24 14:46:10.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"3240b133c997826f7c9a49ea588dcbd4c516f3dd","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-24 14:47:14.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/00f057b8330645d7a9409df64d9f3228 : NODE_FAILURE Node request 200-0000672701 failed in 0s","accounts_in_message":[],"_revision_number":14},{"id":"58618963d5de094fa0694e4da0121ed5ae0666d3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-24 14:53:25.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c3ba0ed1a44a495fa08773f3835a78dc : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0aa87f48283945038bbbcc9148b3fd6f : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":14},{"id":"c07363f8d6efdc43eb320f1684467be105c7c78c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-24 14:58:58.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":14},{"id":"b54ff38499e475b89dfd46165b6a05b63573b954","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 15:08:05.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4750e0985cc4a11bfb007dea2ce01a5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c9b64ba85f74d67b8abfef6af97767f : FAILURE in 20m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7a2a95b515684c509cbc4fac4ae62e3e : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/24c4505999b34ae4b1889aa95dbd4328 : POST_FAILURE in 16m 59s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"1c6f08c8a9fc9faa350c6dc1608c32ec23750dd1","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-24 15:49:41.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver build/5ee1bb247c8846278dbe3bc4ca6ae5b6 : FAILURE in 33m 22s","accounts_in_message":[],"_revision_number":14},{"id":"31b3a6c0ab52089ce7e191301325290673efcdb2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-24 16:13:57.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 38m 45s","accounts_in_message":[],"_revision_number":14},{"id":"a3fa515366ce40102ee7c2d9f6b7c3cf03c4aaae","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-24 16:19:09.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/e93c4e73611940369fa6e735c4d0abb9 : FAILURE in 38m 59s","accounts_in_message":[],"_revision_number":14},{"id":"1641649a6c109f8f63bc26bf67d4ba819617514a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-24 18:13:58.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3559/.","accounts_in_message":[],"_revision_number":14},{"id":"308df7851b9e6dfa1d6c075463642b6a7f589038","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-24 18:50:14.000000000","message":"Patch Set 14:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3386/.","accounts_in_message":[],"_revision_number":14},{"id":"574ac8584d634bdd5edecdcbc5f464f0d78264a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-24 19:21:20.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"227b8a56e4dfc1f0fcc23a4b52f8302054703265","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-24 19:23:11.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/6aa694e0d90d4ff4bf951ea8085f9648 : NODE_FAILURE Node request 200-0000673279 failed in 0s","accounts_in_message":[],"_revision_number":15},{"id":"6d660b3cf870ca4d00344ee6410b92aff900bead","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-24 19:33:29.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d7ad278857ee40099ba0660262afb6ca : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/632f7e4b62074a0aa15c3740faa92465 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":15},{"id":"99871fc85dd6a4269e2e53bd9d5869c143052c0e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-24 20:00:08.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":15},{"id":"919521fbc125393e2f263c357b8fd60aa19d484e","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-24 20:33:05.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-fc-driver build/7db6ef895d534b3481478bf8f9ef7a6e : FAILURE in 32m 38s","accounts_in_message":[],"_revision_number":15},{"id":"604abad4acf6f65a7f82f743a36936f8d5bf5386","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-24 20:42:09.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 37m 55s","accounts_in_message":[],"_revision_number":15},{"id":"d106385f6356158387d54092e7c92f56b64a2c62","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-24 20:55:59.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/0f2e5d5014fc4eda98d8f3b9ed834175 : FAILURE in 43m 21s","accounts_in_message":[],"_revision_number":15},{"id":"0e15ab6fc4154c4b217840b789fdb35df16cf8d4","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-24 21:14:52.000000000","message":"Patch Set 14:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1206/index.html : SUCCESS in 90m","accounts_in_message":[],"_revision_number":14},{"id":"ac612e35fafbf120989e4dcc4b0ec5ae28aefb50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 21:51:25.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c1ad16bb8d9419d857100ae89c41c84\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8404a4a22016474e968cc5f1070a241f : SUCCESS in 9m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d125ebbf3e04e3f919615653dd1b25a : SUCCESS in 14m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/608e9281baf84cebbd62c65243a1d0f8 : SUCCESS in 13m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03ca430bbf064af1aa0eeb29b5a2ed97 : SUCCESS in 19m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74dd7f2754b24e6a8b672bc9cc22ffd6 : SUCCESS in 15m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/50de3aa9e55c4c7282e5c8078622e648 : SUCCESS in 1h 03m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84a0285bddc74af0b7f5f7a7a72e5c3b : SUCCESS in 1h 09m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa3dec6a38c24861a64e55ecc459410a : SUCCESS in 1h 15m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bd0ecbd3e95489d8a2af08e7bbeeac6 : SUCCESS in 1h 05m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a513ce12863f46e082572b83864ca256 : FAILURE in 37m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/98a2934ce781470a96a9528bdc8eaa6f : FAILURE in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9b42eb683ad48758463fd9e6e36498c : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/53a915ac92b94502869b5bf11ac38f6c : SUCCESS in 6m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0865c72161c24d319e9df4971043857f : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8c2cc52392984db7a76f7b7cfd6a1616 : FAILURE in 40m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2a82c11c752a431cbc6e9a6f4bd9435f : SUCCESS in 6m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d9422a757860455ca173d95dcd7eba14 : SUCCESS in 1h 39m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b7c2cac7ae5436f93d3c2c8b9f951f1 : FAILURE in 2h 13m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fd635ce345924e1da292bd059d22e1fd : FAILURE in 35m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/76997c68a1e14f08b50f535dfe91b6dc : SUCCESS in 2h 00m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eb337148aa5c4a198b03974472c2f64c : SUCCESS in 39m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a3701e8f8d8944289e4f76d0db657d1a : SUCCESS in 1h 19m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1f8882adb7aa4e0398dc4d3fca83ee36 : SUCCESS in 1h 47m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/48e31bdade5042178c287fb7645eb232 : SUCCESS in 32m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aa973281c33c473b93c894cde8aa0591 : SUCCESS in 2h 23m 15s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2818460ee2bd4e8b841cbfb0c1c4f1c4 : FAILURE in 1h 54m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f4ba2838c9514e7ea15aa726b20c2835 : SUCCESS in 1h 37m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38facf31b1934b12b89c5d89850fcc43 : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":15},{"id":"6f837632003dad29579d21142b9323c3d73a678e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-24 21:52:18.000000000","message":"Patch Set 15:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3562/.","accounts_in_message":[],"_revision_number":15},{"id":"79e07e1e1870ad8be43aadf7ad26f86eeb9b416c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 22:04:58.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d85c3b7fb7f44dbeb9476f919aeb395c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/604faf34f7a6488a986896a4a8741674 : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/353d4a3014d6423fa7d8eea864c892d9 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3bee042f0e424a3abb41e06de4a99bba : SUCCESS in 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"fc99f51f7bd03652113c109f1de3ade807e436e2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-24 22:13:29.000000000","message":"Patch Set 15:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3389/.","accounts_in_message":[],"_revision_number":15},{"id":"748a01e80ab62744a9889bd8593d95d4f09d7700","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-24 22:29:15.000000000","message":"Patch Set 15:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-959383-15/FC/ : UNSTABLE in 6m 1s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-959383-15/iSCSI/ : FAILURE in 3h 7m 35s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":15},{"id":"6c1a81f03b11863dbeb8fae4b88e31e8294b97ab","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-25 01:55:21.000000000","message":"Patch Set 15:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1210/index.html : SUCCESS in 91m","accounts_in_message":[],"_revision_number":15},{"id":"c93d431af682462f376105623f8ea866cd8c0c32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-25 15:16:27.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"b67762a4fd519cf96e61651487970eb103789d6d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-25 15:18:05.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/79a04ce6ca2b407eaaef763caf873065 : NODE_FAILURE Node request 200-0000675743 failed in 0s","accounts_in_message":[],"_revision_number":16},{"id":"218b4dfa6a02f60047e023a07cd2ef700c1417c8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-25 15:28:32.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bda6016cab2d4bc7963fe6e90ffe3ca8 : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/73606bfc01c7403b832f4e23343606b7 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":16},{"id":"ffe52a2b7a10013035a014507d002f6b7244fb58","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-25 15:35:42.000000000","message":"Patch Set 16:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3567/.","accounts_in_message":[],"_revision_number":16},{"id":"27a25a8959241f2626281611e539b15ebd05312e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-25 15:44:03.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"2a21f9acb46ef52f3bef48eee99ada0d175e1b69","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-25 15:57:37.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":16},{"id":"28d98e71055b68fee07c6fe313b78a8e0c63ad0b","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-25 16:22:13.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- hpe-fc-driver build/870bed07643c4196bb8d44abb6e7af23 : FAILURE in 31m 40s","accounts_in_message":[],"_revision_number":16},{"id":"d23919eee56696b24fbc95411485d4568b3f1d6f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-25 16:36:56.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 37m 58s","accounts_in_message":[],"_revision_number":16},{"id":"2a132df908ff5e6f4b587d857d7f86a15b149171","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-25 16:42:17.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1e27a7e7dd494d40a43789153d4aec7e : FAILURE in 38m 31s","accounts_in_message":[],"_revision_number":16},{"id":"3ae07a1f138af06526c21e970c3e0e5668a39604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:57:11.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be672192228f45f895834ced93acd0d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d22d1778fb054665bcf95cd38784d833 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d17bf73e9abe4db396895402b12fda95 : SUCCESS in 8m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0272978d738d486cbaa0b6615ca64a22 : SUCCESS in 8m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c332c3f265d4ffaa3bad7f80c26a617 : SUCCESS in 14m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4554d70093a04215be5da09b32ac0780 : SUCCESS in 11m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/2cef740b26aa4cd78442ffa14fd02c08 : SUCCESS in 1h 07m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18515813a688491cb2552d4f9c2f1711 : SUCCESS in 1h 10m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6f5fcad91934fac8f4a17580276ae88 : SUCCESS in 1h 26m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/799b2d9a48434580a6aff2f412e7cf79 : SUCCESS in 1h 03m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d604b1e9d30842d9871af3ff42ab4c0e : FAILURE in 23m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/11acda38be114606822e1a169a914025 : FAILURE in 3m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1a94c4de09154e828c936070c562ebc1 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4f768910a0c047cf806d87f0e309a94a : SUCCESS in 6m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6c2032e2a5df4d29af84462d4de1756d : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/65d0c894a5a24ed585f35097217289a1 : FAILURE in 41m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/41b762afea2d498c9e665f751532a5e4 : SUCCESS in 5m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/79582d856aed4e5e8a49fd5663286204 : FAILURE in 12m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2bb089ddb85d4aeabb818dd70ddb82fe : FAILURE in 2h 35m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/685187805fb943bab6dccfd9d4ee2f0d : SUCCESS in 1h 54m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/025b2d7e8f754f3aa7643597b0353fc9 : SUCCESS in 1h 22m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/57ae40f5096449399c548981d1ba43f0 : SUCCESS in 41m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4042356114094405a2ce8e843fc2b351 : SUCCESS in 1h 00m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c39b246220f04b1fb4f2de5c9f9884da : SUCCESS in 1h 41m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b50c0b99413c49cd80478d9e53865027 : SUCCESS in 37m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6d1c976ea7f04b739231474c9bd5fd44 : FAILURE in 1h 52m 58s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3a623050fc774a9caca6a9f3186d2f34 : FAILURE in 2h 03m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/09ea3788f87848349d9608fb6f86ec7f : SUCCESS in 1h 39m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7da883d31ff84a3aabebf7b72bace67c : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":16},{"id":"cef93e9f8a7fb8742cf7ff388413286cc8ee8c12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 18:31:58.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5cbfe70932944aab5496490cef49b94\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9fa63ee7d73e4d8896108f9984f2b007 : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f2c0c3e8162471dbad3329faf318af0 : SUCCESS in 16m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/36cd6d31a4e741efb08d01f52015fd9f : SUCCESS in 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0b643c7886489c8e984ca30aa2b61ffd72c9093a","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-25 18:39:36.000000000","message":"Patch Set 16:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1215/index.html : SUCCESS in 99m","accounts_in_message":[],"_revision_number":16},{"id":"2c33f7dd9fe55195807af63cfb245d70243371d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-25 19:36:30.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"77de0318e367fa946c1b35dada3ccb8aabb79a6f","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-25 19:38:31.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/5f88a24b3ea846fcb82e92f7a9565c94 : NODE_FAILURE Node request 200-0000676303 failed in 0s","accounts_in_message":[],"_revision_number":17},{"id":"d3d51d5ab239fcd512151b1d921cbd493ff43e18","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-25 19:48:16.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bf9cd87c483e4c18bf88511a77a7875b : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1da86b47409a4a868585b439152ccdca : RETRY_LIMIT in 37s","accounts_in_message":[],"_revision_number":17},{"id":"68392dbcc3d66e6deba293547c6331496b49e595","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-25 19:52:54.000000000","message":"Patch Set 16:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-959383-16/FC/ : UNSTABLE in 6m 12s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-959383-16/iSCSI/ : FAILURE in 4h 36m 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":16},{"id":"c4732d0dbcc5efc1dca9eef01aefa4e731d32eda","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-25 20:15:47.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":17},{"id":"afa4786bb85fc3fcbc50f0a1e4de5a4ed26bfb26","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-25 21:36:49.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- hpe-fc-driver build/785f5d69e61a4dfaabcb008c97c16241 : FAILURE in 34m 12s","accounts_in_message":[],"_revision_number":17},{"id":"ab6f50426007e6f84bb9d2478cb1386ff537a937","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-25 22:00:41.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 39m 44s","accounts_in_message":[],"_revision_number":17},{"id":"0725316a7e4835f54f5e729dee51732766f7467a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-25 22:07:10.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/6cf92268ad87468c9d32566c34da8ed5 : FAILURE in 39m 26s","accounts_in_message":[],"_revision_number":17},{"id":"c27fdad164b9dd4a0356785e5209ed941bafc618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 22:10:27.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/726d2fddbbe94e4da5c8212010d59bf7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a38f31fac6b4fd088e036a31dd8822d : SUCCESS in 7m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b7114110e204ca3b660e6b45d276389 : FAILURE in 13m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe9fd92ea7354d99b202584841cdf0af : FAILURE in 10m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6b433db88e6f46f5b6932839b5493276 : FAILURE in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/356043f81c1c46bd99889afe06a4794a : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/7661b2db269e4c80b39312c0fcbed101 : SUCCESS in 58m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f19e3cd5e5d848aaab178f527e28ddd9 : SUCCESS in 38m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b4e4755949764a12829cb3af7f94069e : SUCCESS in 1h 50m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d60932f0a9f464a8ce67cee96fbbaa3 : SUCCESS in 39m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5c4d6d4a0dc445b3a08986e5b7dd3b3d : FAILURE in 19m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/57d024665a7f4a499ce602441d86e336 : FAILURE in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9344a644dda142db9bbc493db42071d9 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/522367dcddbc498996b6187e130fa519 : SUCCESS in 5m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2f35ad179d664528a928bb9babd4e4cc : SUCCESS in 4m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b19c1c476b2a45a7b96dcd44fcf89f37 : FAILURE in 39m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/12135a80fa4b43a08e03f51b3b6923a2 : SUCCESS in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f1b00c7ce2b4479b836cf154adac5657 : SUCCESS in 1h 12m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/10a6bddfd295406889b4739b97f2e6ca : SUCCESS in 2h 28m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4fa2a4b233d6491dba0d7cafb27215bd : SUCCESS in 1h 53m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d81347975e71457d9c47f6f3e408319c : SUCCESS in 1h 56m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1df3892df30e4bc2bafb25ebe6e18e0b : SUCCESS in 26m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d1723d55df124c4bb7d9c5bdfd385858 : SUCCESS in 1h 15m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c2d7271709524534b095bfa481571baf : SUCCESS in 1h 44m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0bfcfcbf18694ec68b44c928cab1e01e : SUCCESS in 27m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3fd94db96727441dbfb7724cb4632815 : FAILURE in 2h 09m 11s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/b117881145b84c929e779d6319ea3439 : SUCCESS in 1h 13m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4aee992d407f49699e986cb76cedb814 : SUCCESS in 1h 40m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/445766de787743809a58368907698d94 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":17},{"id":"7e3804706cd5476fea63e9b94b49b47d86b3d7d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 23:30:05.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75aa75e5fdeb45a3a5a3fd570d3cc603\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01a073415d23474a9fd88b0442eac80d : FAILURE in 18m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3b479739cfbf41be9669ff731f716a08 : FAILURE in 17m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6f0030d73d7d4061819a39610fe04977 : FAILURE in 15m 01s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9c136e866de8c0f0fc79bff5f9bf40a81e067ad3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-25 23:45:19.000000000","message":"Patch Set 17:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3575/.","accounts_in_message":[],"_revision_number":17},{"id":"627e7c8fd545c638a45a07e51597cb2d372a06ea","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-26 00:42:27.000000000","message":"Patch Set 17:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-959383-17/FC/ : UNSTABLE in 43m 41s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-83-959383-17/iSCSI/ : FAILURE in 5h 5m 32s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":17},{"id":"db904f2d12e655184fea78ae732846f9c388bb5f","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-26 06:22:33.000000000","message":"Patch Set 17:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1221/index.html : FAILURE in 79m","accounts_in_message":[],"_revision_number":17},{"id":"a31e2f0e0aeed63407d81954633d853d6dc51a3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-26 21:30:19.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"d9db17c89b2e0779e5ec4b1f5e4342b08df64b2a","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-26 21:32:04.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/3b88aed8728545b085324ad8a621300d : NODE_FAILURE Node request 200-0000679565 failed in 0s","accounts_in_message":[],"_revision_number":18},{"id":"a29aaae3f506eb2404aba3a2dcdc61a4c2950add","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-26 21:32:10.000000000","message":"Patch Set 18: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"b89b69c7a163fd331f3d8c67c315668696a6ef4b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-26 21:42:30.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0a3e34b019f6487eb0875f4386703ee1 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c0797c3ea23943d28cad597f2851c927 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":18},{"id":"ffbd7a7c63703e424ab7b2aa8fb47cc67fab5b51","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-26 22:07:49.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25 .","accounts_in_message":[],"_revision_number":18},{"id":"6f5316a47a061ba5c03d587e4c7d835e61e8ba74","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-26 22:49:11.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- hpe-fc-driver build/dc6397be198d487a9ebf829e20e3628b : FAILURE in 32m 25s","accounts_in_message":[],"_revision_number":18},{"id":"d578508f2b0f1aa53211f9952b211f332421c9a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 00:00:53.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/044ff458b0724ffbb46bb6b3f668d45e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e53cf15c0e814966aed90389bd44a08f : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e599cbcf2716421cab53d4cf646322f8 : SUCCESS in 8m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8492f6ddc3d74cc9a1af95d8d79d5044 : SUCCESS in 10m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b2a9f3858d19445d9b55a9c595b5e508 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7bebe2012547d2b3a97b570cbe003b : SUCCESS in 14m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/a5825c9626434e34aa54cfee18508120 : SUCCESS in 1h 09m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea68ea3907d94f3794fbe224a647a101 : SUCCESS in 1h 02m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4f654cebb4e2423a9177bcc1f79fe6db : SUCCESS in 1h 53m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aba297c3c56a4e1daced9160241d9292 : SUCCESS in 1h 04m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4d27ebba3c564199b26fdd76b7e2ae50 : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8edf18eda2ed49eb9fb84fccdb205a56 : SUCCESS in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2ab2a5959954dcfba5e1f23522c7cd5 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4dfc2c85bcc244fc890a142475c2cf96 : SUCCESS in 6m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0b67118bc02c480fb1af8c86ebf75490 : SUCCESS in 6m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6f464ab63d074b85b74eaa4f2e6d1b11 : FAILURE in 39m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1d497eab7ce543c19848172e8137a62c : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2863d4e3c5c427cb821594b03de8738 : SUCCESS in 1h 55m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6805a0ab16754ecc86617150b92bb9a7 : FAILURE in 2h 23m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8fc53a071a264d24bff9dfe2ca367555 : SUCCESS in 2h 06m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6de467f0ccfc4ae5acbd0d69b3a5419b : SUCCESS in 1h 52m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1188dcc1372d4b23a3b25a32c47e1abe : SUCCESS in 19m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eaf63da57c3a4b40b61fd7476e5d3a8d : SUCCESS in 1h 16m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac620bdc3571445e940d0178673443ec : SUCCESS in 1h 42m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/474e853cc3254fe088d4bb0ed54a7f14 : SUCCESS in 41m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/66b12b2292b049d7a352c11a63a202c8 : SUCCESS in 2h 00m 47s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/93401d1204d147ebae3a5003cc7e9ca6 : SUCCESS in 2h 06m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/507ade1ab4794b8887393d2aaef06cca : SUCCESS in 1h 00m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32a35f5626ed4dc4a55deade4a2a6a65 : SUCCESS in 48m 51s","accounts_in_message":[],"_revision_number":18},{"id":"d4526d832d1ee7ebf548e148f64011c9270a3ee4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 00:08:28.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d39a7120351c4996bfb66a5e428e68f2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/efa374b9d9434063a05620a26eec29d1 : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2b5d14bd467f440c82067e4838a9b362 : SUCCESS in 14m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2a09c02147924e9c8d50aedcb1fe10ea : SUCCESS in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"8b72115b3459ca244a404eeb0a25fa6f91641eb2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-03-27 00:10:19.000000000","message":"Patch Set 18:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/27_03_2026_00_04_06\nLeave a comment with","accounts_in_message":[],"_revision_number":18},{"id":"bd0292085903dac6aea5baedba646d5a158afbd4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-27 00:35:10.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959383 : FAILURE in 40m 24s","accounts_in_message":[],"_revision_number":18},{"id":"3fe6bab11edc2c018691ee6c906d3f27b454559a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-27 00:54:29.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/33480840b0a243c8a1bc9062a84a6836 : FAILURE in 40m 42s","accounts_in_message":[],"_revision_number":18},{"id":"422441bda211d837e783c09417b86bdffaa11f03","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-27 16:11:01.000000000","message":"Patch Set 18:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3609/.","accounts_in_message":[],"_revision_number":18},{"id":"02a1647b99061a790c5e79d962599e7c86eac827","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-28 13:47:48.000000000","message":"Patch Set 18:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1259/index.html : SUCCESS in 91m","accounts_in_message":[],"_revision_number":18},{"id":"03d3996f051cedf76c4f34fd45af427f927caad3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-31 17:35:14.000000000","message":"Patch Set 17:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3402/.","accounts_in_message":[],"_revision_number":17},{"id":"36a6e7e4fe2cc03c22a6479598b767caec969579","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-02 09:57:29.000000000","message":"Patch Set 18:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3436/.","accounts_in_message":[],"_revision_number":18},{"id":"76313cd80743d8fdc17c758037985dbc0d1f5c09","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-07 17:12:24.000000000","message":"Patch Set 18: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"e69455fa89bf2a53a8600b4f8dfcd27775aba63a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-13 13:22:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"7cf05b0b7eb0317de745072e5237d61eae7daea2","revisions":{"37f645ae8f07bfcbe26723e4f7969f7a63405ff7":{"kind":"REWORK","_number":1,"created":"2025-09-03 19:32:52.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/1"}}},"commit":{"parents":[{"commit":"be53d538abcc6c5f45bf8c6016bfc711908bd5b4","subject":"WIP Prototype running with threads instead of eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be53d538abcc6c5f45bf8c6016bfc711908bd5b4"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-09-03 19:32:26.000000000","tz":0},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/37f645ae8f07bfcbe26723e4f7969f7a63405ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/37f645ae8f07bfcbe26723e4f7969f7a63405ff7"}]},"branch":"refs/heads/master"},"dd7cb1e508db6105f4ba73247bf09972f3b12956":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-07 16:38:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/2"}}},"commit":{"parents":[{"commit":"6e6fdb9bd0c293dd18246a04ae799eba8cf2d2a3","subject":"WIP Prototype running with threads instead of eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e6fdb9bd0c293dd18246a04ae799eba8cf2d2a3"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-10-07 16:37:50.000000000","tz":0},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd7cb1e508db6105f4ba73247bf09972f3b12956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd7cb1e508db6105f4ba73247bf09972f3b12956"}]},"branch":"refs/heads/master"},"33dad71e96b77039b1c436defcf49b213aa88c90":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-09 14:20:29.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/3"}}},"commit":{"parents":[{"commit":"258c274ec5dc5fc5eb948038e585423b5d987367","subject":"WIP Prototype running with threads instead of eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/258c274ec5dc5fc5eb948038e585423b5d987367"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-10-09 14:20:08.000000000","tz":0},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33dad71e96b77039b1c436defcf49b213aa88c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33dad71e96b77039b1c436defcf49b213aa88c90"}]},"branch":"refs/heads/master"},"db62b9960e972920f88ac6545ab0faee7fc08824":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-11-19 14:14:11.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/4"}}},"commit":{"parents":[{"commit":"990b25149e3780e2822edaadf36e4a007cf04c52","subject":"WIP Prototype running with threads instead of eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/990b25149e3780e2822edaadf36e4a007cf04c52"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-11-19 14:14:11.000000000","tz":0},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db62b9960e972920f88ac6545ab0faee7fc08824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db62b9960e972920f88ac6545ab0faee7fc08824"}]},"branch":"refs/heads/master","description":"Rebase"},"c321182149e8ed9ed42c6e2bd81ffca141ebba22":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-12-09 18:19:06.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/5"}}},"commit":{"parents":[{"commit":"995a66830adaa722b2100a84279206b27fcbfd43","subject":"WIP Prototype running with threads instead of eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/995a66830adaa722b2100a84279206b27fcbfd43"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-12-09 18:16:59.000000000","tz":-300},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c321182149e8ed9ed42c6e2bd81ffca141ebba22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c321182149e8ed9ed42c6e2bd81ffca141ebba22"}]},"branch":"refs/heads/master"},"dd9a6b95f026a4325d3d5af87dd8711b8b1725d4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-01-12 15:21:25.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/6"}}},"commit":{"parents":[{"commit":"23792b46d746123f13ca69713549b6a8385ff19d","subject":"WIP Prototype running with threads instead of eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23792b46d746123f13ca69713549b6a8385ff19d"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-01-12 15:21:25.000000000","tz":0},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd9a6b95f026a4325d3d5af87dd8711b8b1725d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd9a6b95f026a4325d3d5af87dd8711b8b1725d4"}]},"branch":"refs/heads/master","description":"Rebase"},"a599bac4019cf711d1608c9f74867959f7144e55":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-02-18 15:20:54.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/7"}}},"commit":{"parents":[{"commit":"a8d262e99e16b4ebbaf535d05ec2d446206e1385","subject":"WIP Prototype running with threads instead of eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8d262e99e16b4ebbaf535d05ec2d446206e1385"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-02-18 15:20:54.000000000","tz":0},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a599bac4019cf711d1608c9f74867959f7144e55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a599bac4019cf711d1608c9f74867959f7144e55"}]},"branch":"refs/heads/master","description":"Rebase"},"ca2e1b1ed00bee8269d38fd5d56b01ccd49f4dca":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-02-18 18:14:43.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/8"}}},"commit":{"parents":[{"commit":"2f96498e88e7541d5333c29bfc1193c4e92eab3c","subject":"WIP Prototype running with threads instead of eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f96498e88e7541d5333c29bfc1193c4e92eab3c"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-02-18 18:12:34.000000000","tz":-300},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca2e1b1ed00bee8269d38fd5d56b01ccd49f4dca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca2e1b1ed00bee8269d38fd5d56b01ccd49f4dca"}]},"branch":"refs/heads/master"},"7b98de8e6e9b90d58203c63f663e3056f725d97c":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-02-23 17:46:41.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/9"}}},"commit":{"parents":[{"commit":"31d4f10183c63ba9fe718a4575e7c31b74dbc9be","subject":"WIP Prototype running with threads instead of eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31d4f10183c63ba9fe718a4575e7c31b74dbc9be"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-02-23 17:45:14.000000000","tz":-300},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b98de8e6e9b90d58203c63f663e3056f725d97c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b98de8e6e9b90d58203c63f663e3056f725d97c"}]},"branch":"refs/heads/master"},"1d0a3698fa76fe3e919998d7687c458b1788b3ab":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-02-27 20:02:52.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/83/959383/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/10"}}},"commit":{"parents":[{"commit":"ff6f5f2b37bc78f9c45cec3845190b416eb93edc","subject":"WIP Prototype running with threads instead of eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff6f5f2b37bc78f9c45cec3845190b416eb93edc"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-27 19:25:21.000000000","tz":-180},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d0a3698fa76fe3e919998d7687c458b1788b3ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d0a3698fa76fe3e919998d7687c458b1788b3ab"}]},"branch":"refs/heads/master"},"bce6fb4587049735457537c731342ce904bde897":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-03-05 20:41:41.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/11"}}},"commit":{"parents":[{"commit":"6e3c44200779ef4dbeed6e7a7f1c3c6cf4c17fa1","subject":"Enable threads mode in cinder-volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e3c44200779ef4dbeed6e7a7f1c3c6cf4c17fa1"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-05 20:07:36.000000000","tz":0},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bce6fb4587049735457537c731342ce904bde897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bce6fb4587049735457537c731342ce904bde897"}]},"branch":"refs/heads/master"},"463b6083c30ff7da1f68576fbbca47bcde457f34":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-03-10 17:42:03.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/12"}}},"commit":{"parents":[{"commit":"1c274b9bafc3bd17df3295dc3120417918559b3e","subject":"Enable threads mode in cinder-volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c274b9bafc3bd17df3295dc3120417918559b3e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-10 17:41:49.000000000","tz":-240},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/463b6083c30ff7da1f68576fbbca47bcde457f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/463b6083c30ff7da1f68576fbbca47bcde457f34"}]},"branch":"refs/heads/master"},"fd20686b4d55816d684e8d2581ec3caf21970fed":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-03-19 20:43:51.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/13"}}},"commit":{"parents":[{"commit":"95561639d003ad44e1c4bdc02d973cb396f83ff9","subject":"Enable threads mode in cinder-volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/95561639d003ad44e1c4bdc02d973cb396f83ff9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-19 20:42:08.000000000","tz":-240},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd20686b4d55816d684e8d2581ec3caf21970fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd20686b4d55816d684e8d2581ec3caf21970fed"}]},"branch":"refs/heads/master"},"a0ebaa7500aac015220cc26b029fe9df0f36cb80":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-03-24 14:46:10.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/14"}}},"commit":{"parents":[{"commit":"ea52085ad6a43dd5265c96f0354070cf7d42208c","subject":"Enable threads mode in cinder-volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea52085ad6a43dd5265c96f0354070cf7d42208c"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-24 14:46:10.000000000","tz":0},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0ebaa7500aac015220cc26b029fe9df0f36cb80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0ebaa7500aac015220cc26b029fe9df0f36cb80"}]},"branch":"refs/heads/master","description":"Rebase"},"54bbaa5ea383e90973d60d8209179a2e83690955":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-03-24 19:21:20.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/15"}}},"commit":{"parents":[{"commit":"bbe8e6f36900db700080679c3a6df6be4a40e5e7","subject":"Enable threads mode in cinder-volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbe8e6f36900db700080679c3a6df6be4a40e5e7"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-24 19:18:19.000000000","tz":-240},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54bbaa5ea383e90973d60d8209179a2e83690955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54bbaa5ea383e90973d60d8209179a2e83690955"}]},"branch":"refs/heads/master"},"5f3bbef47f29860a7dd24d35c65ae7ecd9a18cc8":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-03-25 15:16:27.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/16"}}},"commit":{"parents":[{"commit":"ab07eae6f1afe726b01375e61faf07007d5f6f33","subject":"Enable threads mode in cinder-volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab07eae6f1afe726b01375e61faf07007d5f6f33"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-25 15:15:03.000000000","tz":-240},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f3bbef47f29860a7dd24d35c65ae7ecd9a18cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f3bbef47f29860a7dd24d35c65ae7ecd9a18cc8"}]},"branch":"refs/heads/master"},"4ed0f5f8fd6c9ea7906c44da5afce78b44999014":{"kind":"REWORK","_number":17,"created":"2026-03-25 19:36:30.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/17"}}},"commit":{"parents":[{"commit":"cec91e64fde17041c04be3855d2e39e95d6cad0a","subject":"Enable threads mode in cinder-volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cec91e64fde17041c04be3855d2e39e95d6cad0a"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-25 19:36:18.000000000","tz":-240},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ed0f5f8fd6c9ea7906c44da5afce78b44999014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ed0f5f8fd6c9ea7906c44da5afce78b44999014"}]},"branch":"refs/heads/master"},"7cf05b0b7eb0317de745072e5237d61eae7daea2":{"kind":"REWORK","_number":18,"created":"2026-03-26 21:30:19.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/959383/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/83/959383/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/83/959383/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/83/959383/18"}}},"commit":{"parents":[{"commit":"173e0aff1f01685bd3ce2b90ef17715a8eabe657","subject":"Enable threads mode in cinder-volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/173e0aff1f01685bd3ce2b90ef17715a8eabe657"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-26 21:29:11.000000000","tz":0},"subject":"Replace _add_to_threadpool with native threads","message":"Replace _add_to_threadpool with native threads\n\nChange-Id: I3251a2753ee57ce4fe01f3d0bfb2cbeb6d465f25\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7cf05b0b7eb0317de745072e5237d61eae7daea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7cf05b0b7eb0317de745072e5237d61eae7daea2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
