)]}'
{"id":"openstack%2Fcinder~959388","triplet_id":"openstack%2Fcinder~master~I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7","project":"openstack/cinder","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-03-18 15:48:44.000000000","reason":"Change was submitted"},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2026-03-11 19:28:02.000000000","reason":"\u003cGERRIT_ACCOUNT_30555\u003e replied on the change","reason_account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}}},"hashtags":[],"change_id":"I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7","subject":"Remove eventlet.sleep calls from scheduler","status":"MERGED","created":"2025-09-03 19:32:52.000000000","updated":"2026-03-19 00:12:08.000000000","submitted":"2026-03-18 15:48:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959388-eventlet-removal","meta_rev_id":"f2dea14d4e35113a14c71c9c433f4fb614e70cfb","_number":959388,"virtual_id_number":959388,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-18 15:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-03-17 16:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-03-18 13:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-03-11 19:28:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-03-18 13:23:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}],"CC":[{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 19:43:26.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-03 21:40:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-03 22:17:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-04 05:24:44.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-04 06:26:29.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-09-04 07:34:52.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-09-04 08:47:27.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-09-04 23:28:01.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-08 01:54:49.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-10 11:52:10.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"CC"},{"updated":"2025-11-19 17:27:59.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2026-02-18 18:25:20.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-18 18:25:22.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-02-18 21:50:06.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-02-19 17:43:43.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-02-19 18:08:27.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-24 01:28:47.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-03-10 18:01:49.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2026-03-11 05:07:12.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-03-17 13:17:16.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2026-03-17 16:33:30.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-03-18 13:23:30.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"c3e7889870b30a13e068cb67c77ca753782d211f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-09-03 19:32:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb9fd0a0715ed5e0361e1ff6d0ed4cf2a2afb643","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-03 19:43:26.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"19c6c29f4bc3aabda80cb23eaba16c70176e5126","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 21:40:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55ae8e65259f417399a9b86ccee44f10\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aad5b170893149c4b13414618a8d1693 : FAILURE in 10m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/db64d925e03045f19b182eb50fc277e6 : FAILURE in 13m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b1e0475c59734e59a89023ccac04221e : FAILURE in 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed7838ebee107f41a25b27358d1f8428083b0da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 22:17:56.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96e146f41a10415599efe300c915a555\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a1d0671e5f040ca8bef1af17ec93e11 : FAILURE in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1011bc364ba412fa29edcce5156eeaa : FAILURE in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9881bdbdcb8495bab6605517b2793eb : FAILURE in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1047ad1e16934a2ebf2adfbefefc125a : FAILURE in 9m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6bcb87d0b2477b8562650cc5cce8d7 : FAILURE in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/7f018c6936f24955aa0fb05095d9c680 : SUCCESS in 44m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8afff6ac9abe44469a9eddd6c663c722 : SUCCESS in 54m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88defd2963234f0cbea65932ba285aa9 : FAILURE in 1h 01m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2956a563cc5b45a5a0812351f3f4cc9f : SUCCESS in 50m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1d8f7032147248dca1f22b7a208083e6 : FAILURE in 6m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0594c47a2c204dc1ab8f8dce15d0bff0 : FAILURE in 5m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a4381b192c694a08977acf5950df4789 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ef6a3ede0b6d4c74a26fc2ef91d2950b : SUCCESS in 5m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2921244cb7024c88ab5a9e3639eab662 : SUCCESS in 10m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/af1799eeb26b41f29ec0c3022af6a39c : FAILURE in 29m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c997ea4828a4496bb75a13265041d7fc : SUCCESS in 3m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b838f5f58c7840eb98b15d2a3f3f8335 : FAILURE in 2h 02m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f686364e61c04074995c72c19674bdab : FAILURE in 2h 39m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d282cd3bea414b9daf44a3c78d34f1e2 : FAILURE in 2h 23m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e58dcc3fc78f46189d1692ab493b6550 : FAILURE in 19m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/63c51c525b124242bae383fc7e143630 : SUCCESS in 33m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2313e4161e794e1d9a4d101005bf4fa5 : SUCCESS in 1h 14m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a6c9eb4ae6c34d5fb12d1414f2dd1f8c : FAILURE in 1h 45m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/da7a573c40274133bf0c32ef6ef2cfbe : FAILURE in 43m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1bbcb4cb84f34773876cc918f4710f99 : FAILURE in 1h 13m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/29b4013086034fedaf912b62ece915b6 : FAILURE in 13m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0f52f653ee4a46e89dd7e5a03166123f : FAILURE in 1h 34m 51s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fd8bf6194d24b8b84881c2f236522dc : FAILURE in 1h 20m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6362c86e1afd4f7a81fe79dcebc2ecd0 : SUCCESS in 54m 07s\n\nWarning:\n  Comments left for invalid file cinder/cmd/volume.py\n  Comments left for invalid file cinder/service.py\n  Comments left for invalid file cinder/tests/fake_driver.py\n  Comments left for invalid file cinder/volume/drivers/rbd.py\n  Comments left for invalid file cinder/volume/manager.py\n  Comments left for invalid file cinder/volume/volume_utils.py","accounts_in_message":[],"_revision_number":1},{"id":"edc9f0f7f979e0d4f358dc495e70b98aeefb5f0a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-04 05:24:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f242512eb91541e7968b35254a5c0cd9 : FAILURE in 1h 15m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4374c08c858f4f2694811755ec6e077e : FAILURE in 1h 19m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1346f51a5edf11cd1f7b81ad86a5e2584d826981","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 06:26:29.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/438/","accounts_in_message":[],"_revision_number":1},{"id":"68a3f409ad999879355f7904a13673082a0bc504","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-04 06:26:31.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959388/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959388/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"77f284be288e991f35dc117fbe6fa1b2d9be69c3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-04 07:34:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : TIMED_OUT in 3h 00m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0d9339000e519bf4db72098444ef72b771daee55","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-04 08:47:27.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1672/.","accounts_in_message":[],"_revision_number":1},{"id":"0fafc18fce12a57344bc476605a41d6f9d594c3e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-04 11:11:54.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1808/.","accounts_in_message":[],"_revision_number":1},{"id":"43495aa427ebe102889c19fccec8c94f44090ce6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-04 23:28:01.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"31caa44516b5069c7b61a64c509790f704cac7d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-07 16:38:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0ccda6af1342e6b683c172cf5fc3f9e66c8cdce4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 18:01:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa247f5b9a6460c824e15437fa37a89\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cd4d9e2fe3a942f49044acfc50ce7167 : FAILURE in 10m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aaadf4cd79624d0fa54aab570228bf28 : FAILURE in 8m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5eb701937a9a4c4c80e63a6fc302824c : FAILURE in 18m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c7949ec6840b986b1fd92d7032a5a5bd9ad7d64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 20:02:45.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/930bdf48b6594b4c9765c845f8374948\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99677f9e7da3403d9c581f607ecfc358 : FAILURE in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1dc031ddb3d6408387e835962d5ffe18 : FAILURE in 6m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bddd38f3eda5447ea8b1780413b0895b : FAILURE in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cb6c601d0b647b8898a9bdf9c6ea808 : FAILURE in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e385f7492d4444d8ae4d24877cc23469 : FAILURE in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/ea232705cce14874ac7928c2b6093d16 : SUCCESS in 54m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/13e4e557d2c64de493b0fbbbc2f30355 : SUCCESS in 59m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/39e34624b33d44cbb810a3c424852cfe : FAILURE in 1h 52m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0daf95a1050441749b797185b9a0ef08 : SUCCESS in 55m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/278e793f6ffe4ecd94d1e747fee6ba68 : FAILURE in 3m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9fc579acedac4fd5b841bc9b0d481e86 : FAILURE in 4m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f66f70c54dca4e7ab2d116c4bb4b81fa : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/002873a32d044430bc5df084e5d5e036 : SUCCESS in 7m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/56a7e1b3efcb40c5b3085c12533b705c : SUCCESS in 9m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/50288f68f70e4bcca1658bfc575e7c7f : FAILURE in 34m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/23fbca2a6f9a42d09a0e651bac865380 : FAILURE in 7m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cb68bada273048e796d588d53f48dd9f : FAILURE in 1h 18m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6e5e34265f0546b58a807925f3769dce : FAILURE in 3h 03m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/75d4de17eb86468b9294a6fbd01b2875 : FAILURE in 2h 30m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5c1f627c37ee453896570993023e39d1 : FAILURE in 2h 27m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1e49d580b4724beab44de62814db9cd6 : SUCCESS in 34m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7f18775e48fa400e9299649460bece66 : SUCCESS in 1h 02m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0dfabc86aa71432fb958adb5642f6cb0 : FAILURE in 1h 33m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8ad9598616b34f109ea5d85a68b7b36b : FAILURE in 45m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2c26ca49c4774b91bece2fcf7d0f2c56 : FAILURE in 49m 14s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/6bfa11bc42cb4d78964d0f5fbb8f2134 : FAILURE in 1h 06m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/32c7df0ca8b545bb9ccd76a36bd8844e : FAILURE in 51m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5866f283a6ac4d55aabce549ab2c6930 : SUCCESS in 1h 01m 02s\n\nWarning:\n  Comments left for invalid file cinder/cmd/volume.py\n  Comments left for invalid file cinder/service.py\n  Comments left for invalid file cinder/tests/fake_driver.py\n  Comments left for invalid file cinder/volume/drivers/rbd.py\n  Comments left for invalid file cinder/volume/manager.py\n  Comments left for invalid file cinder/volume/volume_utils.py","accounts_in_message":[],"_revision_number":2},{"id":"f3c302c0c226007e34f65f44edf94f86a52f7e7a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-07 20:54:37.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":2},{"id":"09809b413eec16586890291af3c93c1fd1236844","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-08 01:54:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/959388 : TIMED_OUT in 3h 01m 21s","accounts_in_message":[],"_revision_number":2},{"id":"0d2269829f6de1ebec80fb9a5224d60d6fa9de83","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-08 02:45:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7e2f1348fa5a4075b4227036eb4e75c5 : FAILURE in 1h 15m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/00978a63ea604d2fa0462cd3d003c902 : DISK_FULL in 1h 29m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7468e9f3385dc7d643f5cc8825b5beb11d0f2869","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-08 03:15:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : TIMED_OUT in 3h 00m 46s","accounts_in_message":[],"_revision_number":2},{"id":"253ceec154eae43dcefe174414280660ce05eeb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-09 14:20:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"df0fd99cf3ffa3fffe233d7fc5673d9e90bf8519","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-09 15:02:26.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":3},{"id":"48b098e0599952f708ae4a0d599fd240d43292a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:11:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fee11ecae48d48718eda44b44a1561cf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1415f861e9f24283a98fd5cdf78a264c : FAILURE in 5m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c410d7cc0ad400797a372e29d84fb66 : FAILURE in 8m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aca75cce496a4731946a845880e4fe4c : FAILURE in 10m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d40ac3cc0a3d4921e43464be417a9e8db935b80a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:24:06.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71bfa01647c64905a817e3da35305a30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74891f187c5a435e836b799377d4beb6 : FAILURE in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4e219464c26471684546fd5960b353e : FAILURE in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7e1e2fae6f04868a9d9be9bf4b1a9f8 : FAILURE in 5m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfeeac7b235347e8980d3a9273abe2d8 : FAILURE in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc0a94aa89b4ed9aa1c16928d9701a4 : FAILURE in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/251b54b168244e45868ceda6a95e25a1 : SUCCESS in 56m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b6db523c6384225a42058c220f079ac : SUCCESS in 52m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f07533552ce541efbd5df7963399674e : FAILURE in 1h 20m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31094233a64d482f84ee9cda85efd88b : SUCCESS in 1h 03m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/30c4cf4280cf459caaba88d7b0243737 : FAILURE in 6m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9a2cd856ae5f41f0894afbf9a0530cd9 : FAILURE in 5m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5f6a9e3acf540e498420fa3d5b1ae52 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/efc387bcde9e47d4895cbb0d47d5316e : SUCCESS in 6m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cc426d44398846aa8d821087c58a019c : SUCCESS in 7m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3aa2430565ed4a9199540b19d4b51f1e : FAILURE in 45m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/641da6aec0fa40efb40fff980d759a33 : SUCCESS in 4m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/860a55eaec904419b048f5fc602fb257 : FAILURE in 2h 01m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0cdb0f64e511492da7d5e5b316f8a209 : FAILURE in 2h 56m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5861e45befb34714939b7428c2aefe23 : FAILURE in 2h 16m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c405c0fab61a4159bd00b6f8905804bd : FAILURE in 2h 44m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2d0e5cb0ec054d37a65adf7a3c1cd15a : SUCCESS in 36m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/04f0491d192740408f24af1bfe0220fd : SUCCESS in 1h 08m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e420bff2b4834599a9bbf1e3327886ba : FAILURE in 1h 42m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/78cb170aa123481aba4ed3be24d9ddbb : FAILURE in 49m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/90670fde019d4cd182541c3de300529c : FAILURE in 1h 28m 43s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/77a11c183bb44b4e822366f8a6e48436 : FAILURE in 1h 34m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9aed2fed721a42b39ea3ece8af17137f : FAILURE in 1h 08m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/058987a5d4204b50bb5679f5be34d395 : SUCCESS in 58m 42s\n\nWarning:\n  Comments left for invalid file cinder/cmd/volume.py\n  Comments left for invalid file cinder/service.py\n  Comments left for invalid file cinder/tests/fake_driver.py\n  Comments left for invalid file cinder/volume/drivers/rbd.py\n  Comments left for invalid file cinder/volume/manager.py","accounts_in_message":[],"_revision_number":3},{"id":"93be2aaaeaaabc85832d6b72f56803c276b60902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-09 18:31:56.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"787dfa426a3ccb135ff742f30ab4b6f403919dbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-09 18:46:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e08a9936e5bae831a35bd9d11476023adb0a5412","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-09 19:12:44.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":4},{"id":"ed655ea283e7b481e4d1fc35f71baa1eb1fe1a65","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-09 19:28:02.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":5},{"id":"747dd0c05ce9b21da9b7a1223529343bd9a83d30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 19:53:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97d0d6cb1ed940729130e0b5395e54a9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f8eb323be37246299c581364389f488b : FAILURE in 10m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c464b5dbb07f48a0961b1d2e5c83f711 : FAILURE in 7m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/95071ea91a9c4ff0ba166bf745b4aa6b : FAILURE in 18m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fffdc0848fd9204d313f58c69ea474680a2e159c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 22:00:47.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c65c6412bf734539944599f7ac003c94\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1577165dc4e94f4ca74d9d3e2801e900 : FAILURE in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9407f558935c40298658e14825e0674c : FAILURE in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ef4b1af7e554f4b898f30ee1cdd20c4 : FAILURE in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d573b60ec9b48e9b1f8b537c9e34b9f : FAILURE in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbaa056c00ba4b238e0b8a54fa933466 : FAILURE in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/5dc0be6f3c8248a7bb24d2616912eac0 : SUCCESS in 1h 00m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5aea0927241647ac83340c7fce1c93e1 : SUCCESS in 1h 04m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e54e7887030144bda9283b58e951e0ba : FAILURE in 1h 28m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4be37be3f2549c0935bbc6d4fa08a1c : SUCCESS in 1h 04m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a41534339dff4d2c818ebf02fd176b52 : FAILURE in 7m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/18e2fe3f93eb42699fb065ea67406ae2 : FAILURE in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d6cab9f892e549fca403fc1ad63b060e : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62f379c9388e4b57a8c937a2fce54bd8 : SUCCESS in 6m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6d1e97231549469f99b3d4699939340d : SUCCESS in 9m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e9e2ef39c0d14a539413b300011385fa : FAILURE in 38m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bcb1b8ac559a4771b910bd360fc355ce : SUCCESS in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/72df63050422462f8bc236d486bd5799 : FAILURE in 1h 50m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/23984b7273ab433c8ca284dae212f695 : FAILURE in 3h 02m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8939f5648a1d459d9501ad39d3211fd1 : TIMED_OUT in 3h 03m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1c90b83b0c2048c79047a7cc0d934d4b : FAILURE in 2h 49m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/38a7c506402a41cf812969a206df5bee : SUCCESS in 23m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8fd1536314f041c68e70cd91fdc40d53 : SUCCESS in 1h 14m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5fb50bbfe575476a8f80c26cdc5f9d87 : FAILURE in 2h 05m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9a46697331944130a536f31a7e4fab2f : FAILURE in 36m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/551ae694a8684828b406a4e74e14e5a4 : FAILURE in 1h 23m 07s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f90ba6f982ea473093f74d01e4e8c0a3 : FAILURE in 1h 27m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/195b29c1818c45a3b58bba2ffb595e8c : FAILURE in 1h 57m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed7f391abf3d4248a646b1d3996641b7 : SUCCESS in 34m 32s\n\nWarning:\n  Comments left for invalid file cinder/cmd/volume.py\n  Comments left for invalid file cinder/service.py","accounts_in_message":[],"_revision_number":5},{"id":"adafaa3ede741fe6f71ac21aee9ee93d993a5ab6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-10 04:38:07.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d55b2668672c49aa96c768c422f18ae0 : FAILURE in 1h 24m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/687dd0c0f2534b98a351bcff921ee3f9 : DISK_FULL in 1h 24m 22s","accounts_in_message":[],"_revision_number":5},{"id":"486eaf7a1e421ae719eaf7093c7a9cbe44fa864c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-10 06:10:59.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : TIMED_OUT in 3h 00m 46s","accounts_in_message":[],"_revision_number":5},{"id":"8069f1e4f2ebfe8fd2761ddefcc7a8af1cc2df41","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-10 07:24:34.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1974/.","accounts_in_message":[],"_revision_number":3},{"id":"c348e1703a02db2802bd6962edfd7d68ff7a1774","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-10 07:38:58.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2088/.","accounts_in_message":[],"_revision_number":3},{"id":"4f7ed1e82d9a49c4916bcdfcb86ed550c54b6221","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-10 08:13:06.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"534593856eca6ce91202dceb4be356f59e10669c","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-10-10 11:52:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"69a6db485a7f8fee4ebfcba2dbbb98238a507233","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-10-10 11:54:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fc09ad3c1ab724f00e5208aa7061bf849ed76b0f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-10 13:22:05.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":5},{"id":"1c1b76ef3f300bd578e8166090947743cc56a661","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-10 13:55:52.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1979/.","accounts_in_message":[],"_revision_number":4},{"id":"5d8ad72996b9e9feda07b44db80f92ba2bd6eb6c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-11 00:23:49.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1984/.","accounts_in_message":[],"_revision_number":5},{"id":"3b8b4b0cecddedffd3e074374caba1decbfa8e8c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-11 00:47:38.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2098/.","accounts_in_message":[],"_revision_number":5},{"id":"e48c89f7d33e4b5f374a564c1b49887f99364d2a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-16 19:39:52.000000000","message":"Topic eventlet_dev_4 removed","accounts_in_message":[],"_revision_number":5},{"id":"df3481ff8af51c20e5de39b625bfb8cbd7db41a5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-16 19:39:56.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":5},{"id":"b72244d9385a32563e195d8cdf9e1df95e0d7478","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-11-19 17:02:44.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1e70a92d6d92ea319bac6b312fda87d4925a7068","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-19 17:02:58.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2517/.","accounts_in_message":[],"_revision_number":6},{"id":"f8b9ed8b8b793b2f4fdfb8ab6726f870c1565fef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 17:26:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48242733fed7491bb80a4d2ea95050af\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/541015e52a99478391cad912dc0bbe0b : FAILURE in 14m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/63795fd3a4204689b1923dcec46c8fc2 : FAILURE in 7m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/953367010b2e47d0a944b43e80788239 : FAILURE in 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8779dbcf106b01355c4a0fe3464f36d48839b397","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-19 17:27:59.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/13/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"ab90d62e476bcc7415eab38cec7efeaa59ce3325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 19:59:16.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/adc5d8cb81e64fceb5300af24f8e34c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ffe853233c4aeda7d6cd518472d51c : FAILURE in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/708b72365a374709b3e4eeed193192b2 : FAILURE in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c3e36d677174638b57e2c4169a53279 : FAILURE in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314fbc5b3ce24e53a7e9c18c4da54ff5 : FAILURE in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/70117157aaad416da1a3fdb9b6b11e00 : SUCCESS in 30m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fbd0867c63274b31888ce59f02702150 : SUCCESS in 1h 00m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f870495602a4fc9b376cc2747a177f2 : FAILURE in 1h 26m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf2e930065d147ef99bbe692b883b28f : SUCCESS in 33m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2f2553305c81495a8fe48fb32902f4f9 : FAILURE in 5m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6db44d5003ca4a9497dfa8a42da75eb9 : FAILURE in 5m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f2bc623f1df4db2bca4fdb1d656fb92 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f6d9aba02b240e58c9217df1543bcc7 : SUCCESS in 6m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0d31588548db4d72b815da4ce1dd3fcd : SUCCESS in 7m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a624f2e8b8e7479eb34fe04693a1205a : FAILURE in 32m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f365ad7909c4b719746b2da320d80f0 : SUCCESS in 3m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6630be40e7b84a1e8e17ee55dc0669f9 : FAILURE in 2h 06m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a80586ed85e4f25a965cb3863725091 : FAILURE in 2h 02m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d054ec79874a4816aab1e06411b4be9c : FAILURE in 2h 54m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f7dd565539d64090aa8cf22d5ed93e0e : FAILURE in 2h 20m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/96088ddd3b774f81a8059a8113c39f7c : SUCCESS in 35m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/994501666a6540ca849d76186af3b01c : SUCCESS in 55m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0b10c5c656fb42b6a564d0f5df94551e : FAILURE in 1h 53m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/788f0d6cd1d14fe3b64ece2ce43e9071 : FAILURE in 43m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0380d05e038f448da2f5e26c48b91a2c : FAILURE in 1h 54m 46s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/56fd8cd5f7dc428f8d171254411fdff3 : FAILURE in 2h 43m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0be12370c5b64c2586abf268bed382af : FAILURE in 1h 43m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11b9ff8c6ec842a88c9177671f291e89 : SUCCESS in 30m 06s","accounts_in_message":[],"_revision_number":6},{"id":"2672ce655a0c920ea309dc12ba93bd7829473ce2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-19 20:28:53.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":6},{"id":"b9f2d261a3e7fc60d4551cb7f0ac59be5cca28ea","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-19 23:49:45.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c936ddb69f6d4f01aacc7227c3fd5439 : FAILURE in 1h 05m 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e70ad85f01524f56ae2fbd58dffe89dc : FAILURE in 1h 08m 34s","accounts_in_message":[],"_revision_number":6},{"id":"e595059d4e12fba87416ca2191305c7c3226f858","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-20 00:54:24.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : TIMED_OUT in 3h 00m 46s","accounts_in_message":[],"_revision_number":6},{"id":"a660cc0cd6499e14a373f3fefa8845388499fe39","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-20 05:39:08.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2661/.","accounts_in_message":[],"_revision_number":6},{"id":"fccb22d57a852161351575daa9e86a89280676ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-12-09 18:19:06.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"e8a54e30ad33abf401dd278637255db14d7e5949","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-12-09 18:27:15.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e2527e5bfbeffb360c17aefe3bdfdeed35bdac56","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-09 18:27:16.000000000","message":"Patch Set 7:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":7},{"id":"9a9d71cfd0fbe42dc7f2bb9084ee0185e9262c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 18:27:20.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/963618 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"2ad0f5ed63158e849c827449f6b9e2558980fb33","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-09 19:22:43.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":7},{"id":"bfae0443c8bb5ea470dee6e3614bda340684b006","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 19:49:18.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/041a1c316e584824982d839b0a2bf7a8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec63123150d741c19aff143f7405b677 : FAILURE in 36m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bc54d716fc72408793d4c2276d7198c4 : FAILURE in 13m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8fcf1c24305b4fafa79fa7ecb181b383 : FAILURE in 26m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7c5f9a7d1524dd79e86160dc77bd812ad17a6a05","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-09 19:51:01.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":8},{"id":"d903df9bdaf621456f62dfda78cb980c4d1b9c52","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-09 20:05:35.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/164/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"4674da5442825a6e0298a0a054a193eb676230cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 21:27:31.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1261da85a6a644d5b455bf3c5d912c09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35db6d5b5495478988c65e6f0a54100b : FAILURE in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cab6b5e2fcaf457180d296d18a8fb3e7 : FAILURE in 12m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/162d102e200d4464bc18a3616d16f1c7 : FAILURE in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea9e83968d31434ca136ac45f2ced27a : SUCCESS in 13m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/5fae65fc64d644f497d6657268c24e42 : SUCCESS in 29m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4054dc023eea43e786a4d84b6fa9bacb : SUCCESS in 57m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d772c518a07a4d818fbb9c09d04bb776 : SUCCESS in 1h 21m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93069de9ed1e456ba400f8abe60b3455 : SUCCESS in 35m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dbc665c84cce41dc9e4ff43aa0a16d1a : TIMED_OUT in 40m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/80202253e6d344459d34d8a1bfbef523 : FAILURE in 2m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/41dd348b54a44dd89b896a23858f618c : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f36d68c4b1764c0c8073dc7d26f14faa : SUCCESS in 5m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/306d328023534e7bb779a85f608d456d : SUCCESS in 9m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2cf20e01a390425c9db8b09a4305f4db : FAILURE in 18m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/376f946d6868463bb777036fdc86131f : SUCCESS in 4m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8e4c10cfa1b94b1696ded63618d29867 : SUCCESS in 1h 50m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3e2442d77e36458d97fcf7a419e6c3f1 : FAILURE in 2h 47m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3dda8adbe17c40a6a07a293bf128729e : FAILURE in 2h 57m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e8cd9b0adc50478ab4f1cb7d0372fe38 : FAILURE in 2h 15m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dc042f51259a4a93927c850cb850ad01 : SUCCESS in 36m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/401c1e52ac384fdab9195fbf1e5933f7 : SUCCESS in 1h 03m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/08f2c2ba2fb646d79dd86decfec0fcb3 : SUCCESS in 1h 22m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/066be912bc804b379ea3ed891ef45374 : SUCCESS in 43m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d85bae743f424d6f9703a6c5c955876b : SUCCESS in 1h 04m 19s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ae61ac33e2a14e9fa103c60aa4269f90 : SUCCESS in 1h 58m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b61b7bc89c4f4ef0a26934e722f7076b : SUCCESS in 1h 42m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec833a8053e64bdf88b5e32cce42dae2 : SUCCESS in 51m 36s\n\nWarning:\n  Comments left for invalid file cinder/cmd/volume.py","accounts_in_message":[],"_revision_number":8},{"id":"5c7a171bb7b72c2d2e4f9efec9bc6ffed49ca6da","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-10 00:47:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2814/.","accounts_in_message":[],"_revision_number":7},{"id":"77b462b04dd8bd6d2251afe61329533ef2f5481b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-10 00:51:47.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2667/.","accounts_in_message":[],"_revision_number":7},{"id":"56af2ecf5d04d2c110194697bf0e9a003ac73925","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-10 01:48:55.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9c0ad053caef4580b430a812566bfb73 : SUCCESS in 44m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1d254b5c18674ddca1ad2951de72bc66 : FAILURE in 58m 12s","accounts_in_message":[],"_revision_number":8},{"id":"02c53b6de8a1189445c2883db4287804d6f81f50","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-10 04:13:29.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2818/.","accounts_in_message":[],"_revision_number":8},{"id":"6ba98acc121017aecfee29b902f8401480d6e07d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-10 04:17:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2671/.","accounts_in_message":[],"_revision_number":8},{"id":"a81e0a1d116341a0571bf02220b7b884247f6a38","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-10 04:38:47.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : FAILURE in 1h 05m 45s","accounts_in_message":[],"_revision_number":8},{"id":"26a00e96f313c7cb471f810e6fbfe50346bc2fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-18 15:23:24.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"1dc9abb4ef38f6cc49ff6c2636a263d45a59a7e8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-18 15:49:16.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a236d27e2c8c405f826acccb44bb8a7a : RETRY_LIMIT in 43s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1a584232cdea4e6b8727883248859c9b : RETRY_LIMIT in 44s","accounts_in_message":[],"_revision_number":9},{"id":"c7b2ca525794394565224a0440c3849068a3e3ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 17:05:28.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60ec578cd6ef427aa8d6484b601b7d70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2beadcc7e1db43a0a52660605c192e35 : FAILURE in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e525f809591b4868bb12b0b033db6c62 : FAILURE in 6m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e2adb76426e4357ae590271088a2644 : FAILURE in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d95711da9e4b0ba2a86ac746eb0b75 : FAILURE in 5m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/1a8495b0c3a64d7d9823aba047824541 : FAILURE in 30m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d67412928d654e6ab766caca0ef0b698 : FAILURE in 51m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa5b0d405a1a43f090f438a5ae70f75a : FAILURE in 27m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/266158417e4f4c54af126d1cfe166a8c : POST_FAILURE in 24m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5069589f2e2c4c9484146c6f33dd2017 : FAILURE in 6m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a24f0ccaf4b64aaea5db2aec258f0757 : FAILURE in 6m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/099b5b2c18b643ed825a0f44cb4627ef : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d9c7a8a3401b40d3a402080dfc4de7e2 : SUCCESS in 6m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bd23ec2339ae432b8d3eadac4736d19a : SUCCESS in 6m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6aad3ef8640c4ae2add5eca0fe6ba3fc : FAILURE in 30m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a8b49f559eb24ac1bd270e3dd317cb31 : FAILURE in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d72f808680cf4a85bbd4e7c540b497c7 : FAILURE in 32m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9754214dc6954174b30a9dbe0afb96a1 : FAILURE in 34m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0d4de22e321d4025b2f67cb216ffeeee : FAILURE in 28m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1925f4e26d274511908fc53f45081a66 : FAILURE in 39m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a4742ab49c094cf2a9157abd71aea6e0 : FAILURE in 12m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a0b96694e7384ca3b37bb6633618d734 : SUCCESS in 1h 14m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b5abc7f2f3f24fccabef267c2ebe392c : FAILURE in 30m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/19ca7dcde7eb47a0ad5dcb30042f5121 : FAILURE in 12m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2689e686fd524ae581c58814328290f6 : FAILURE in 28m 17s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bd82a0408b364710906ae0af45d91953 : FAILURE in 26m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d20165245d4d4279aff69fd6a144bfaa : FAILURE in 25m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6800baeca4ac42638f8087fa2b9ce12e : FAILURE in 19m 26s\n\nWarning:\n  Comments left for invalid file cinder/cmd/volume.py\n  Comments left for invalid file cinder/service.py","accounts_in_message":[],"_revision_number":9},{"id":"f434e39aade80adbc336ccae0be475d90a477037","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-18 18:25:20.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/963618, https://review.opendev.org/959387, https://review.opendev.org/959386 are needed.","accounts_in_message":[],"_revision_number":9},{"id":"64d6bfef2f2f6bfe86460f079293a28030f3ea34","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-18 18:25:22.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/963618, https://review.opendev.org/959387, https://review.opendev.org/959386 are needed.","accounts_in_message":[],"_revision_number":9},{"id":"6f2275bd215b94f810926f233d8c51412b1e595f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-18 19:14:56.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : FAILURE in 38m 47s","accounts_in_message":[],"_revision_number":9},{"id":"565618e01d01b1d0006f5626cb1024a19b1aaa2a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-18 19:16:54.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c32468c8034b490190406ad5d95976d7 : FAILURE in 40m 08s","accounts_in_message":[],"_revision_number":9},{"id":"8414c868415b11cd7f129d789f2bcdfc8a12319c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:29:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92131f33890e4b82b857466dda13053d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/227dd61c16e2459a85eafa467fd9aeab : FAILURE in 9m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/242bb689f8e847309f2aaf5b0c073dc8 : FAILURE in 11m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dfa95bdade0041d9b0576c240d3a4808 : RETRY_LIMIT in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6d7f4db48740a0fef691b714134053d619172374","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-02-18 21:50:06.000000000","message":"Patch Set 9:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/732/ : ABORTED","accounts_in_message":[],"_revision_number":9},{"id":"be85899ab14624bba3074bc156758747bb4dd1d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-19 15:38:47.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"44486afa8b16c2942c55aa11b3c46484277fc95f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-19 15:49:36.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6e7e03d63b3b44009d0cce281a6ddbff : RETRY_LIMIT in 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f13e2b30dc1a4a06868b08bcd80e9543 : RETRY_LIMIT in 52s","accounts_in_message":[],"_revision_number":10},{"id":"274a1bf3e47ba26059ebf4cf7d58e2a6c630aeca","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-19 15:50:04.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":10},{"id":"932355360cadecacd95c9d7a3f1655f1888990cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 15:58:24.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a9dd403534449d3af245f1e65ee1ca5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d8ac8c0fc684813bd03c21c983f9b05 : POST_FAILURE in 11m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4b141fa85554a1db0951b63e4f510a6 : FAILURE in 13m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4d4c69d3a1964effa66e7495a58dbe79 : RETRY_LIMIT in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9d88ad60eada0361a93417948ad6654c4cba2b0f","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-19 17:15:36.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver build/82640e3d0cc744acbd01aec383d17044 : FAILURE in 43m 36s","accounts_in_message":[],"_revision_number":10},{"id":"272ef4990bc0767e7d4b0ac446144cbebc5983bf","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-19 17:43:43.000000000","message":"Patch Set 10:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/959388 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"13ed282ed39198bc43dc4e349d435f689c592b4b","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-19 18:08:27.000000000","message":"Patch Set 10:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959388 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"7723c5ed7a7bcf473578cecb819de9c634f5ba84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 18:30:24.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1ae1d36ce5e465bac2fb0fe745dde86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2222e2279c34c8b8d85fd5c3b72193b : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39fb2a02906d4a8c88b72ada9962298f : FAILURE in 9m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18ff5b01da1c47a990751b20c2913a15 : FAILURE in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc1a65909ea24f119bd36bb8dde07b38 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/28f8cdb5e74641e48378feeadb6cd5fb : SUCCESS in 32m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c44431c703f49caba402b99f7d514f3 : SUCCESS in 1h 08m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3c4a12e44234a45af821ea53a15ef53 : SUCCESS in 1h 24m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0631cef034f47f2a0d1f28c10d07e3f : SUCCESS in 1h 01m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2a2e49c2da0a43b58412c15df130a2c9 : FAILURE in 18m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4cdf811f9f244324a9ccc970b95a7ede : FAILURE in 3m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/368d456a9dfb499bb7f1f7e540756a6d : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e273a22735344814a0845f4ed5926ae0 : SUCCESS in 4m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/639a9d24fb3648db823560771c1fcb39 : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/96252eecc17e45e8aabda6bfdd132524 : FAILURE in 41m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f2272d2064a5440b83608a5fcf917094 : FAILURE in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dea3ecca0bf04f8fbcdc0a011b1846a1 : FAILURE in 2h 04m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4cedec0493dd4ff0a25bd7f97207f9f9 : FAILURE in 2h 15m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ad0425338c70481b84e43252de03b3e9 : FAILURE in 2h 07m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1dff319ecb124402a1aa0bf2673846d2 : FAILURE in 2h 48m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d9a1faff93a547f1ae903e94f4aad61d : SUCCESS in 40m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9b8ea2c4bc7140749cb18405a821a64d : SUCCESS in 53m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dbd6637e96494bb6a8fc204f9cbe3d54 : SUCCESS in 1h 33m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/adcd4c08bcb84dd7a1db4225cdda72d8 : SUCCESS in 41m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1d07fa321dc442eda59af68fd3d37796 : SUCCESS in 1h 31m 11s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d78f6638d93e4afc9555f035fe3b3414 : FAILURE in 2h 06m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86afd47b56ea4c4bac1082d37e32c50e : SUCCESS in 1h 46m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8eed8d948f314a52a4c4844db17cd9c2 : SUCCESS in 43m 01s","accounts_in_message":[],"_revision_number":10},{"id":"3d2f80c675048655b771687ac4cc1fdf93e630ba","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-19 18:33:09.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/14d6efd65ce04d07a8ec54bf4d32d69c : FAILURE in 41m 09s","accounts_in_message":[],"_revision_number":10},{"id":"e7d6fd0031ca55993e6b97c6909043d44b2ac022","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-19 18:33:15.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : FAILURE in 38m 55s","accounts_in_message":[],"_revision_number":10},{"id":"c657d1bd09b1d0aec98d9f03597754e5c2c2ac35","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-19 20:01:47.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/b55dd809d07f47af98a44de5a4313e91 : FAILURE in 45m 37s","accounts_in_message":[],"_revision_number":10},{"id":"4eb70742ef01619a5aba1dddaa2a3a2d5f28dd21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-23 17:49:48.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"45a805ef2ec250a4e71dc8dbe25c15755b4c1a49","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-23 18:29:34.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3a6db8af2cbf49e493dec8e758992767 : RETRY_LIMIT in 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/318edfafac0843a88f0d6e28c51f1a32 : RETRY_LIMIT in 1m 12s","accounts_in_message":[],"_revision_number":11},{"id":"bdf2f7ebfe664b1c3effbbdd5f7eef5f78f0ce8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 18:55:51.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a975db62ed7f4c8aaf21b93fca728e12\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a996e34ec72409b97059906ffcf8683 : FAILURE in 30m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e33f455ede142a28ff6be796cf9caf9 : FAILURE in 22m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/24bfa551afb240d3af1e2d44fbaff8d4 : RETRY_LIMIT in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"899e8b2b27cd0e3afe6dc91bc31651d67a9b444b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-23 19:05:33.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":11},{"id":"5314a699d57c8bf291a79e1bb74e919bf30e25a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 21:20:35.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa604b5cb2e1405f980272bcfc22097e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60950ce600ec467590f0cd1434197c96 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c65bf46e396647618cb8e2d59569d237 : FAILURE in 13m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d14f6b2072742218ad40b9636e905b2 : FAILURE in 8m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6175c491d6bd44718bb7578f52f9fcdd : FAILURE in 15m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea2fe077c8a84026881f6819a0d88fb3 : SUCCESS in 14m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/9f44ac3b54eb4f0689930f2787c4635d : SUCCESS in 34m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1fb733594b9c48988ee4fb0ade07d99a : SUCCESS in 32m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef2cd278220d440d81006239f6abe498 : SUCCESS in 1h 27m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fc43f7cf14f4e8db7a9757f79ea9785 : SUCCESS in 38m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/38a451a4179448f1ac75dd807acafc09 : TIMED_OUT in 41m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/479d008c79a8475b838896650b0c42ce : FAILURE in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5d5bfefaf20d4c859d4c790eca87251f : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ad6f49f50ed4a68939ca7555e687738 : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4eba79c9e60c4d738031327486c24a9c : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/681cbb2114784ebbb4a72e9621510c23 : FAILURE in 41m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fa0145692ecb43689e2638b26266b28c : FAILURE in 7m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6bdc43df1e4b4c0fbd208424d2f4fdc1 : SUCCESS in 1h 52m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b5eba89d12924a268102036ac87cb8dd : FAILURE in 2h 25m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/994c9f01fe904b7ba44c1d131c3d3101 : FAILURE in 2h 44m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d0baccea8f8c4c36aef59634c3f986ce : FAILURE in 19m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/103fbfbb033545a98e95c7bd6daa7473 : SUCCESS in 18m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5f9eba2c766449d7b528010dda55672f : SUCCESS in 1h 03m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6ad9e7c43bee43f0b6281592a224b96d : SUCCESS in 2h 05m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4959857e91c1492494fb85933667c65b : SUCCESS in 45m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/98323108791743e399fc00fd07ee43d2 : TIMED_OUT in 3h 13m 31s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/7561743315184de3883b16af808f5e9e : FAILURE in 20m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c4b44a4d83cf468382c17736bbf94dfb : SUCCESS in 1h 39m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75b088f0bcf240fcaac52a5ffb6bd6e3 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":11},{"id":"24a044d8d99a0cb0ef9c68afb1c4b40e037660dc","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-23 22:48:00.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver build/297d715901894dc3b4a8285f11be42d7 : FAILURE in 30m 52s","accounts_in_message":[],"_revision_number":11},{"id":"f64ff15df8de63457c0cf389b2ec74133d89f847","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-23 23:33:08.000000000","message":"Patch Set 11:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/959388 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"0816f01ff41f286d1ce370463f065446e0b76c55","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-23 23:42:13.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/d621d9d60eb04026b5514f2c33834842 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":11},{"id":"73a06273e577ac9d5a1779e7f73adc9f57dca47d","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-24 00:05:18.000000000","message":"Patch Set 11:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959388 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"d745c93c7e8fae6038a5b95dedd6640c0400f4c2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-24 00:17:03.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : FAILURE in 39m 18s","accounts_in_message":[],"_revision_number":11},{"id":"d6d25dd6dbde2bf48991edf286552ecdd4f72687","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-24 00:34:37.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/0038a3dfd1074f96bec47a286108b481 : FAILURE in 41m 18s","accounts_in_message":[],"_revision_number":11},{"id":"39caafca756b4649611d0457f7658c39ba78b93c","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-24 01:28:47.000000000","message":"Patch Set 11:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_831/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":11},{"id":"26fe9f3969b7d1dda5b2fd79a07adb42becbc174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-27 20:02:52.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"b09d333cf14d0c46155b0a5f23a0cfc4da718cc6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 20:04:24.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3101/.","accounts_in_message":[],"_revision_number":12},{"id":"a90b50c7925f6627d6e0a5f1f5c5754eb00092c1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-27 20:58:36.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/647f0e5cd702460e94c969526798a0c5 : RETRY_LIMIT in 47s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/13ffb8fa880c4621b771b2ce1f87e27e : RETRY_LIMIT in 47s","accounts_in_message":[],"_revision_number":12},{"id":"f3eccd87a684dd9461802320b1d44ec3c3c538e6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 21:22:47.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3265/.","accounts_in_message":[],"_revision_number":12},{"id":"65a933f38734cc57886e387f51429f8976eab1bc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-27 21:26:13.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":12},{"id":"e7070909212696dea6243270ac729f698715366f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:01:23.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcaf8aedf43048508588439a4c42c418\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cba44acea1fb43a39cf260de0782ff76 : FAILURE in 30m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ff94720f96554a5988613afde69431c2 : RETRY_LIMIT in 4m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8341812fbd7e477c9f249c90bfa635b4 : FAILURE in 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e2a193d86f943cb31171c9930744e4e021d1f243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:52:30.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f5ea61b949b4b518fad9b095a83450d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b05b19781de34987ba8f3dd9c4338456 : SUCCESS in 8m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/923f6a051a064832984bde351bb5c48f : FAILURE in 11m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7fa4533010fc44138424c6000b6ebb79 : FAILURE in 12m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f27ff95d94624dc8b6b81f42ee5107a3 : FAILURE in 15m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61144b00602049d4a32f8c26a4f9cd80 : SUCCESS in 12m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/3e83dbed543442a2a24533435426bdf9 : SUCCESS in 56m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5cdd2536d2f24e3185cb7c5a38779fd9 : SUCCESS in 47m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/318e0b43f73944ef9c295ad7913e2e0b : SUCCESS in 1h 50m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bad2797622a34446af20474af75f2042 : SUCCESS in 1h 02m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0a74282569ef48e0bf791f18c13df55e : TIMED_OUT in 1h 01m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/278269777b324de1a99532e445827934 : FAILURE in 4m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0eff65f6ea2744ce99832a2547bb9e25 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f0dfb479bac4897b5ac2b5181f982fa : SUCCESS in 6m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c6026475496f4695bfc5b27cee577597 : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8226b39edcde4222a41336a6eaeb084d : FAILURE in 35m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7966386536a745b7a1e1282d0681a8c6 : FAILURE in 8m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd74e0b03c874ca8933ec5652b028d72 : SUCCESS in 1h 23m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f84f5d14de474f5089f8e16b6da726a5 : FAILURE in 1h 39m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6dce3d52631c4106aa4866f910e0fc12 : FAILURE in 2h 27m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e5c1708ab2d64fbabdb7893246383251 : FAILURE in 19m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/374bd9d375af43889df4eabe9e9a3a2a : SUCCESS in 40m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1e1d0a2f906a4274a300be5963c7b530 : SUCCESS in 1h 11m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4a31cb00b3ec4f92add67a0cfd70cbce : SUCCESS in 2h 04m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/06677643ebe94acdb5a0ca07b311dc8e : SUCCESS in 36m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d3434e6042b14829ac5845b95c30cd41 : SUCCESS in 1h 55m 54s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/baa692c4b38f441cafef3781b3eb8971 : FAILURE in 22m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa3ea6e3949046e8af0283943849f693 : SUCCESS in 1h 36m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c1f5b6aec294599a4be2ebe38f506ca : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":12},{"id":"23919d36f82b7323e276d8699e437e3af6321597","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-28 00:49:32.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/9fde5cd5aeba493b86e7fbb0120805db : FAILURE in 4m 07s","accounts_in_message":[],"_revision_number":12},{"id":"5e0c2e38ced768cc157f91720b2b2489094100c5","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-28 00:53:28.000000000","message":"Patch Set 12:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/959388 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"b5766ae46097e499ea423f9cfa0eab7a16c6ef6f","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-28 00:53:34.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver build/d09637a4c82348f1bc3540f2cb632179 : FAILURE in 4m 05s","accounts_in_message":[],"_revision_number":12},{"id":"11d311c7707b7207cff31e945e6c1aac72f4fced","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-28 01:22:26.000000000","message":"Patch Set 12:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959388 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"d91f27f25b850da7a2a53a3cf8463d52dd880c91","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-28 01:37:53.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : FAILURE in 11m 42s","accounts_in_message":[],"_revision_number":12},{"id":"2f4bc7753b38d75bb37c8cf437283e0062fe0b98","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-28 01:56:15.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1649584736274d7195a1014a0697183a : FAILURE in 13m 15s","accounts_in_message":[],"_revision_number":12},{"id":"827635aaefba9dca13c3b6f8a63240edd272b087","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-01 00:02:23.000000000","message":"Patch Set 12:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_896/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":12},{"id":"73398534e98fa46dceda5aaf472a649c791016bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-05 20:41:41.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"a00ff71e0d496a5474064c3d441055c0fe173b4d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-05 21:18:36.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b83749edb61849358907a5dcfaee3335 : RETRY_LIMIT in 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e933c6e4970d4e1f91d5ef5db6703a99 : RETRY_LIMIT in 49s","accounts_in_message":[],"_revision_number":13},{"id":"f2115213c4c75f13b9dda052b2b19005785ff89b","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-05 21:24:16.000000000","message":"Patch Set 13:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/959388 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"830ac200b42d5e01a634cd3ccf195131d1312e0b","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-05 21:45:07.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/4b07333146844928a28ffe78db97c50a : FAILURE in 12m 32s","accounts_in_message":[],"_revision_number":13},{"id":"9a227ccb204365ce0b7b3a59826cb5b135f428d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 21:49:39.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a9c60250bd54eaba3be94e20c4e61b0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1ceba1d886c54af1947ea005b854b380 : FAILURE in 19m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dbddcefb27874665972213f9c815cb2a : RETRY_LIMIT in 4m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8e2a979a511c4241adeb219e87121d02 : FAILURE in 13m 04s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"76064cfda3648ef698ebc83c0341257bfb60e516","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-05 21:52:47.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":13},{"id":"468521f49504183cd397b15dbb4749a594af75c1","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-05 22:21:59.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver build/a9b5b874433e4d0fb7614924ce25aa75 : FAILURE in 9m 32s","accounts_in_message":[],"_revision_number":13},{"id":"23abedb7a37e7f468707da46d5de1cac1f097117","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-05 22:34:28.000000000","message":"Patch Set 13:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959388 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"ab87f29003cc70f5b53a76eb2420d9a0a9df5c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 23:11:40.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/726dc6c5411b4a49b310d50c85e8518e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0eb89858fbb490cab5733cea2d6e2ba : SUCCESS in 7m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16aab7ef1fff4e2182bc0ab7208ec51b : FAILURE in 12m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bbe63d81c2fc423a97075a69c40b526a : FAILURE in 12m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/00a3ba961bf2411bb2faa1f065ae1e37 : FAILURE in 14m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/476fe7b516244ee3928882ddbb23d08c : SUCCESS in 16m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/aa2fdeb02471448db0be544d30fc0d01 : SUCCESS in 52m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e54e0df20f204d3db8265ee0319c469c : SUCCESS in 1h 00m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/324cdc46b2a34343ab06055706f7913c : SUCCESS in 1h 31m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd0394b9205c456da5231cc96d47c77a : FAILURE in 1h 37m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b2dad3ed50344749849d1b657a8cb4be : TIMED_OUT in 41m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/46351affe06644a8926a6f033a5313d9 : FAILURE in 6m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7159ddb86eca4816a2fcb00c09e7506e : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7f8382c523f447f2b44a5c178511f0cb : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d2940ae2ebed493a86e3c08990a0af1f : SUCCESS in 7m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7e421bc5bbad4176bb02cf29437252b6 : FAILURE in 37m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9dd568ffb0504d728890d012b06f0c38 : FAILURE in 7m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4b632ee4d4d34e1bb7e1ce6dc2a0539c : SUCCESS in 2h 08m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e499bce2ab2441b9be5fabb55290fd1 : FAILURE in 2h 12m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/10740c27473f4330921b26f224f719a8 : FAILURE in 2h 05m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/565e4724b9f64f4c8d4a933b2408e881 : FAILURE in 2h 04m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9de5336278d04d9c8056e2b544d9a5ac : SUCCESS in 34m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a7d73e5a9926495b8f98057478cd877d : SUCCESS in 1h 06m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cb7dbd133749463c92769a32a6e403c8 : SUCCESS in 2h 02m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3058441a39e54cda88d45ee885831554 : SUCCESS in 29m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3aca57a86b1e428abf0b8d5379e1e453 : FAILURE in 1h 48m 50s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2add11c4520e4ae1a0559482fab03b5f : SUCCESS in 1h 48m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5470c71e388a484bac52d7a2217e4426 : SUCCESS in 49m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0ac4c9361fc408fa234d9fc940d37ec : SUCCESS in 32m 56s","accounts_in_message":[],"_revision_number":13},{"id":"2233d61575985d07c0d444d9461ca701b93be8ec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-06 00:07:26.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2f44aeeefa584c31b12781861010dd6a : FAILURE in 20m 05s","accounts_in_message":[],"_revision_number":13},{"id":"3dea00d274518786ebff7c9f7f241698853f794e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-06 02:48:04.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : FAILURE in 39m 29s","accounts_in_message":[],"_revision_number":13},{"id":"65010095bcf3082eabe4c0b042948be9660b1252","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-06 04:41:26.000000000","message":"Patch Set 13:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3179/.","accounts_in_message":[],"_revision_number":13},{"id":"3e038e71978d43a5ec13cf6aafa4bcd66288f836","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-06 07:23:20.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3352/.","accounts_in_message":[],"_revision_number":13},{"id":"5b10af56ce7214c73797ff3513b4738bfa7cab5d","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-06 09:12:51.000000000","message":"Patch Set 13:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_979/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":13},{"id":"feadac624985dcff7461a1580410bbeb1fab142d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-10 17:49:40.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"f58880db2a6b77e8dbb81b2d0bbed689a78c3465","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-03-10 18:01:49.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"306327b9dfc3ca16d10804fed63b37781c78ab2c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-10 18:06:05.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e17f74d3a0d6486db93c1fc5773d7ee6 : RETRY_LIMIT in 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0e0f29b1cefb40ee96b2ab4436d87684 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":14},{"id":"688f72a0378a348493ceb313181e1affc479df6f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-10 18:27:13.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":14},{"id":"b9d218bf0b3fd3a053789e8f6959e39cb33c41d4","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-10 19:00:23.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/2433990bc17043acbf9ff0ca362af97b : RETRY_LIMIT in 22s","accounts_in_message":[],"_revision_number":14},{"id":"97eb91b016c3dfc94bdceb8aa34db262dbf68bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 20:08:35.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd85e615dd7146a5bb73ef9bad867201\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a75f6074c2ed455faaf10dcd12d967b1 : SUCCESS in 8m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06e6c6234a404ff4aa907ffb63a412fd : FAILURE in 13m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b2b37d461774333a0d536e600c289d5 : FAILURE in 10m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/34fcb1b6eeda4c70a91693dda4711f9a : FAILURE in 11m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25042d51be7d47488fbf6a70899a84b1 : SUCCESS in 18m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/e365dc53745e4f978f2915c9e06d1a12 : SUCCESS in 1h 03m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/05dff7a7d3314a588c5056b2299a54cb : SUCCESS in 1h 16m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/582585e42ffd4a8c935a080be0fd134e : SUCCESS in 1h 24m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d02525cebb7246489d460707ab20030c : SUCCESS in 1h 08m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/86c684b7236147619230443905a19f72 : TIMED_OUT in 41m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2601a1acf6654d78acb7382324bdc66f : SUCCESS in 4m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8b5891e6ff7042878f8218346321588d : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1e1c86da7934cb09ce087aa98bf3ad5 : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b844e13bfd264654b32d162dc3b4f9e7 : SUCCESS in 6m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e960e8b20ebd4191bb4c213ea03f01b2 : FAILURE in 44m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1c0f85a763714a3f82aba96b5653fafc : FAILURE in 7m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a9abc80fd09a48a2a133047619e2386c : SUCCESS in 1h 58m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f9846645691e4d439b67acc1415b5afc : FAILURE in 1h 33m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/927426baa22747bf8dbc15e7904bfbe7 : SUCCESS in 1h 58m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c65d1b69d8af451bb2e9c2025b42042f : SUCCESS in 1h 58m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a2ac171d7c1b46a9ad37d3c4c7f2eb21 : SUCCESS in 22m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/60de5f66ba1c46bcadcf1d6ad7c48731 : SUCCESS in 1h 11m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f7d5fd9c01ca43138a3331fe4bc0efbe : SUCCESS in 1h 36m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/62c78abd48464b4590426b3261b59e83 : SUCCESS in 36m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/75933ac130bc44c783d6a0055c2ea363 : SUCCESS in 1h 11m 26s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/35c9001093d942b99b5afc65a63ca09d : POST_FAILURE in 2h 11m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f2fb3ce15bf47d2978ea0a82aeea6fb : SUCCESS in 1h 59m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee3e854d6e0c4819877dd15f8b753f20 : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":14},{"id":"4a4c3a40d1555d945e3c26f462f1df7ed5a5e7e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 20:37:44.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad6fb42eb47e4db788e4a1abec8a6de6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4262c47b5644e79a5d7c88249cbba96 : FAILURE in 17m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f41ed644615c458f922064960147da2b : FAILURE in 16m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/020618884802402a9f2ecf548413baa8 : FAILURE in 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2fd091981d5e5e5c375c77da6546c19346b240bf","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-10 23:08:45.000000000","message":"Patch Set 14:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/959388 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"f00fd2c1291cc3f000b0425c7d8c24d73d34a872","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-10 23:55:35.000000000","message":"Patch Set 14:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959388 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"41941e15dc33867947a86cfad886a2d46ccb7c0f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-11 00:08:46.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : FAILURE in 39m 45s","accounts_in_message":[],"_revision_number":14},{"id":"b11cba9724ec29d4466e26aa2f0eff51472d8c0e","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-11 00:27:37.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver build/ad5c3428939d45c78e0adc509ec5d819 : FAILURE in 34m 01s","accounts_in_message":[],"_revision_number":14},{"id":"4f1bbe348bae3726b02539e630131cf352e7b613","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-11 00:38:22.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1b65ae18cdb74629a9dc1489b809bb09 : FAILURE in 39m 36s","accounts_in_message":[],"_revision_number":14},{"id":"d165e636d34ae3f2d8ab73a989d078d6f919670d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-11 05:07:12.000000000","message":"Patch Set 14:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-959388-14/FC/ : FAILURE in 9h 52m 31s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-959388-14/iSCSI/ : FAILURE in 8h 5m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":14},{"id":"19626d7805ca12921f26f3bcfd3389b79e8475a0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 07:25:05.000000000","message":"Patch Set 14:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3258/.","accounts_in_message":[],"_revision_number":14},{"id":"3de8db47624184d72ec4a515929f518f232ad1b3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 08:35:28.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3431/.","accounts_in_message":[],"_revision_number":14},{"id":"03052cac29f4cc777c34ac8879af5f2360908580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-11 12:26:05.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"d16596cec782cdae9ff077b4cd5f396eb924eecd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-11 12:31:25.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c5de832258854b2981f3d78ed90a3bb0 : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/73645412d82246ca8e133afccbdb9ab9 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":15},{"id":"6d58a0c1cb546e5a068c223a2906351550a7cb84","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-11 12:41:28.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/f0064bff6a5a4d78823733b8e069b290 : FAILURE in 11m 35s","accounts_in_message":[],"_revision_number":15},{"id":"a513ad4fd19425a3e33aa94ea4aab55d9a5f25a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 12:43:56.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd3fd041fbf9429f9a57629bc0db573e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35520efd5510475c910592db9c687fdc : SUCCESS in 15m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f969a36dc17840c6adfec56767441229 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d0fb101763df45aaa9a99f2a638e5b50 : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e0edeef6c72175d6394b878c8534c7093363404d","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-11 12:58:35.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-fc-driver build/c989776e2fb342abbb4097ec31d0aa98 : FAILURE in 31m 59s","accounts_in_message":[],"_revision_number":15},{"id":"a6b66065fe6d91c7a46ee5d9f9536b17478ca92d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-11 13:01:46.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7 .","accounts_in_message":[],"_revision_number":15},{"id":"114802e98d0f0a0209cb340b1eaf2cb82dd51e50","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-11 13:06:47.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/959388 : FAILURE in 39m 26s","accounts_in_message":[],"_revision_number":15},{"id":"cf13a06285258e7017d0e57d73bed3f00a35d77c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-11 13:11:33.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/ced197ac848c47df977c94898c8a9388 : FAILURE in 44m 57s","accounts_in_message":[],"_revision_number":15},{"id":"f0fe1c73727fd93cef73d9b49b27fbab5e1c4284","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-11 13:21:21.000000000","message":"Patch Set 15:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/959388 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"74bc7859f681e44d9ba2eebffcecf9dd1b204a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 14:43:50.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f14e115e498e4496968240ccd9078550\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c459ab23bf0a482e926974cfbb8be858 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad6cfc506bd84e73aab40086ccd7ff81 : SUCCESS in 12m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20fc048200994a6fbace17357be2127d : SUCCESS in 8m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aba81f61c5a74b6d927f71985d6906d6 : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad0cc12ee9214d6dae32fdad6bbebf36 : SUCCESS in 14m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/123dd2ff79634bdabc67e0ac9e185449 : SUCCESS in 28m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cddfca184a8944ae957e7843448614b6 : SUCCESS in 35m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab55c03b3e5f406e9c8c1b383d369598 : FAILURE in 1h 28m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a69e4072597b426c867a970165785ad0 : SUCCESS in 1h 06m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/04146656cf1e4c94beb7180dc79f9f2c : FAILURE in 23m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/019f57fcf5ae4f6b8b36cb749611faa3 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c6975fa09ac64a558f9d3c53853d6ceb : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c3167ea01cc74bbeb598f3ff144043e5 : SUCCESS in 5m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cd83627f5f4b4b17bb441ab12057ea5f : SUCCESS in 5m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c5c41e33d0114e54a6f340a97cebf0fc : FAILURE in 29m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4ac544a8102f41a8bc0b88c7982ec27c : FAILURE in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b6af1e7f9edd44b6bdd65a9fd7dc9133 : SUCCESS in 1h 04m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cb2d655b559b4aecb78d5a92b8bd10f2 : FAILURE in 1h 31m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5bcfeed072a44cdaa52a0ab25c719218 : SUCCESS in 1h 51m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/35d63678c116499a88d1d719b02511da : SUCCESS in 1h 54m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/504e203809084bcb95fa1f5adc3e976c : SUCCESS in 27m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b785cdd8b6a947d9a9e372faeaf44d15 : SUCCESS in 39m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f99d4c42bb2f4b42aa9fa520f72d927b : SUCCESS in 1h 40m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e9f82dc789e04d52ab40089d8bf22c73 : SUCCESS in 47m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e895c8d102a94d36bf00addd3b0ab0c2 : SUCCESS in 1h 57m 19s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/5aa7235034d2496d9326c088a3109135 : SUCCESS in 2h 10m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4041f7884c564cccad3ca75928055dc5 : SUCCESS in 1h 57m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bb82bf30387434c8af26bcf62f583e2 : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":15},{"id":"01054bb10979f033830e990c08dd54ea1e107bff","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 16:42:43.000000000","message":"Patch Set 15:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3266/.","accounts_in_message":[],"_revision_number":15},{"id":"8221db9bdc2473a2e62d64e66c8eaf378b35b684","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-11 17:57:43.000000000","message":"Patch Set 15:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3439/.","accounts_in_message":[],"_revision_number":15},{"id":"072a257b659ff41143a4dc5ff57580ef6282d76f","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-11 19:06:38.000000000","message":"Patch Set 14:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1057/index.html : SUCCESS in 111m","accounts_in_message":[],"_revision_number":14},{"id":"6147658fc1b86a0b31d2dd6582ff4aad75f45bef","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-03-11 19:28:02.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"606b2a2774983820365ca7d741d5304a031c72a8","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-17 12:54:38.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ab3ffadbe64f919b8d9af0e57add97322c75ab11","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-17 13:00:33.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9543fce6c399472e9005221a7fef5bf5 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ae817b3ada294718849f059ea2451f6a : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":15},{"id":"9237d710de891c7a1e9997c2c256b81fcd74d56f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-03-17 13:17:16.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f14cb5e28074b88e1e8c9457cd54936c714d6ef1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-17 13:22:56.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/70f768bda7e24ab78445a37028014571 : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/32ef82c8f5e44d438eb27c2407b0cb48 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":15},{"id":"ee24da3c7546d3672374221f2dc945bb897617db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 14:14:46.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e661dfb2bbed4b69905559522899ede0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ae8e9b14d30043b3a11d4ea9df40bb28 : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2cce8dd0fbc34600becf75ce980f1dbb : FAILURE in 20m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/286d8c6d54af41c3a1e030f2195a74a5 : SUCCESS in 16m 04s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c9f702f026476b0e3d5fba4fef01627b97939e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 15:17:16.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55b5d9502a874c13aca2aa4de215b0a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce46c37f8e33441b82c2bc058331c2a1 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f737d1445a7f43bba089f870e042556e : SUCCESS in 11m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c01703e64775449e90fff8fb8e451539 : SUCCESS in 8m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/adfb55ee6356491298cd3e536f5a766e : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e117914151647bdb42a69f494bbf49f : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/71c19fa9f259456c897e6972bc1a7276 : SUCCESS in 50m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e03279aee8a24ddbafd23543b205c994 : SUCCESS in 1h 06m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/886b3a871d8e4a279fcea42fe2c5a0bf : SUCCESS in 1h 40m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d11b0b9502be4eb8ac50958a90f689e9 : SUCCESS in 1h 00m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6267011ae7034be48e1b165cb04b9640 : SUCCESS in 22m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3f70d77e3be54f40943568754e14dc65 : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7994d55d26774eaab7082bde7c423569 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fb685d46746c4e3bb1aa9a92e2a23061 : SUCCESS in 5m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/28446cb741ab4ad9b486fc11eca52668 : SUCCESS in 6m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8d6ba48a9f90404b8bf7316e79fad683 : FAILURE in 38m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7f2e32bdf0a24367bb3505bea286c4e6 : FAILURE in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b57615e6558640e3bddd1c6b982e7391 : SUCCESS in 1h 39m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e848699415b74f4a9b48af092b6f5164 : FAILURE in 2h 17m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6357c3f94e564d77b15923a286f0531d : SUCCESS in 2h 00m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5097825cf8734b8e9a1d565e8fe5caae : SUCCESS in 2h 08m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5b348955f92946feaa7dd2d3fdfc7449 : SUCCESS in 22m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2f36ba225c484b11bfc55c7fac7fe435 : SUCCESS in 50m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/603c7f53d18647f691a6c5df3a95be8e : SUCCESS in 1h 27m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/44ac2a8d4dfd40429318881988a42c69 : SUCCESS in 40m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9015a2a03b364613a26b66600ce62531 : FAILURE in 1h 45m 20s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e1bb34a2c4084adebbdf16a0ff03203a : SUCCESS in 2h 06m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e3d5f970dcc6415ca9ed9f41172fb025 : SUCCESS in 1h 50m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0517c06e4f8d452795f47d78ca73d451 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":15},{"id":"6d92742c871ecf7d063e646135f015263f30b6b8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-17 16:33:30.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f897e101e6f298c098ee43466dfd78a7f9c71f09","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-17 17:25:10.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"bba96c0c9feac99dad29b50927bc50a4a62ee6cb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-17 17:33:09.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fa34aae831fe4954bd39995916c7e9a2 : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c74ac382c4de472db0d3b41f73070d9e : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":15},{"id":"fb1a43f5c5787078681ccf2f77e8edfdc5608a61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 19:54:41.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2790aee8d42434f87dd60c3cfc11ab6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb69782d809a49a899b2fb0d4d85879d : SUCCESS in 20m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b536a7f260ea4682992612c624d1f093 : SUCCESS in 15m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0d9c2fd9c386484dad7534b2eccb96aa : SUCCESS in 25m 02s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1c9f743a96923f9cf6f8d01912b24346264ca641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 20:09:35.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/659d52b3ad6f472cb1f7f454eab22d4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff66b81376714b7cbcacaa1b1eb44412 : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef6a3d22d3904e91a2430d651e59bd0d : SUCCESS in 12m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3da93a0b7e944ebdbdbbb4a658c33e72 : SUCCESS in 10m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/92e6bcbeaaf44d88b43bfa7500ae5019 : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e62e81d593de4cad812032535012d4ab : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/d9ca920dd07049538bd6381465b7cfeb : SUCCESS in 57m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/470273cf385f4e7a92d69fc61e05e56d : SUCCESS in 1h 00m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d3e7105b6234e5aa1a2473820512803 : SUCCESS in 1h 17m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7dea4bb34c0d486dbef7b519bf775bdd : SUCCESS in 1h 08m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/18fdd6e52ce140448dae82f134dadafe : TIMED_OUT in 41m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6f7cc14947854f5fbecaae9526a792dd : SUCCESS in 3m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5382c5ce3a114d15a20e694e8666458f : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/11b8712b9c4a46588dcf3f2183934bca : SUCCESS in 6m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/597191c306e14e5cacede32a8354177e : SUCCESS in 5m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0af5ad7f43d544078c951122f9c973ee : FAILURE in 40m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe1c13154cbd4aba860699d1f9e36475 : FAILURE in 4m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ecd096d8189e405cb4b132a21746a59c : SUCCESS in 2h 01m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/577e754b849244a89715b704f737f3f4 : FAILURE in 2h 32m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9dadc2063d2741f2ae59572fe022c59c : SUCCESS in 1h 52m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9a23edb688e945c08064e11aa4cf56f5 : SUCCESS in 2h 05m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/af28b2a4fac64e90a39f266c20f1540d : SUCCESS in 43m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ea7dcdca8b464824b9563eac4cab4b3a : SUCCESS in 1h 11m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/839ff022f5cc4d69934d55fe03b3a5cb : SUCCESS in 1h 48m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2686490160064027891139af08bb5f7a : SUCCESS in 45m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8b59b65347284e1ebeb4a5065a2f37b2 : SUCCESS in 1h 11m 59s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bbb147cd2a6e46348825c1e633899663 : SUCCESS in 2h 05m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b8befeaa670e48a7acae4c553a20c8e3 : SUCCESS in 1h 42m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da66b3c142094030bce260acaa16be5a : SUCCESS in 59m 50s","accounts_in_message":[],"_revision_number":15},{"id":"343306fdc46a9d9c65ff0882daf1aa0e727a81d6","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-03-18 13:23:30.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"77bd0b059a31b580192a962611b3de5744fce079","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 13:24:05.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"6e2e276b40c97e8629c1e7476d89d0e18e174fac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:48:44.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8ee9d5089814e2088e7b81f7ecb755d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19b9df5105284f0ebded551efedfba73 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/996d3ecb492d41d8a8a2a8a45d2ba415 : SUCCESS in 9m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ca535de911a47b4bba4f4cea09a7eaa : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3affbee28eb24d9493cc6a1fdc1b0803 : SUCCESS in 11m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/d8aed4f7fcba4da3a70d61af2cc206b8 : SUCCESS in 59m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a46ae58c97ed48e6b5414022c08f355d : SUCCESS in 1h 02m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e0ada79f767408bbe0976934042ad1f : SUCCESS in 1h 59m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a2001674372462cb1dabad3ef632454 : SUCCESS in 1h 07m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d51bf829f7be43bda8c6c2c5d8d42f54 : SUCCESS in 41m 29s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a543c4aea25b46bd85bf0224c348ed30 : SUCCESS in 1h 49m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d69ef9490914f3f8b6fc26be58bf1d3 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":15},{"id":"a973128174ac3dfe12b7bd46765a804085bd0ce8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:48:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"1ab8c59002cb2a8501f28303ee623bec22df8c7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:50:04.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7d78afd52314caf9ebe93e50e18bffc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f12499e288504869a6574d879307e9ff : SUCCESS in 57s","accounts_in_message":[],"_revision_number":15},{"id":"c6f6f256544d7adc1eac417a36a97ccf1a3ad8ca","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-19 00:12:05.000000000","message":"Patch Set 15:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1171/index.html : FAILURE in 86m","accounts_in_message":[],"_revision_number":15},{"id":"f2dea14d4e35113a14c71c9c433f4fb614e70cfb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-03-19 00:12:08.000000000","message":"Patch Set 15:\n\nBuild Failed \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/1171/ : FAILURE","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"caaeae085af95af2dd3ae07cd62a02b2acc8e638","revisions":{"a557a84574447fa50bdafa68392ee54e444bc632":{"kind":"REWORK","_number":1,"created":"2025-09-03 19:32:52.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/1"}}},"commit":{"parents":[{"commit":"d741b884c8aa303450908ef0605600f768111b56","subject":"Add thread info logging to volume manager for debug purposes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d741b884c8aa303450908ef0605600f768111b56"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-09-03 19:32:26.000000000","tz":0},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a557a84574447fa50bdafa68392ee54e444bc632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a557a84574447fa50bdafa68392ee54e444bc632"}]},"branch":"refs/heads/master"},"d16b45969f95fba70c1883f34a420b0594e75395":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-07 16:38:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/2"}}},"commit":{"parents":[{"commit":"60e1649f9d95cc385fd8eea68ceeb9151e21cbd0","subject":"Add thread info logging to volume manager for debug purposes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60e1649f9d95cc385fd8eea68ceeb9151e21cbd0"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-10-07 16:37:50.000000000","tz":0},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d16b45969f95fba70c1883f34a420b0594e75395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d16b45969f95fba70c1883f34a420b0594e75395"}]},"branch":"refs/heads/master"},"c63be9f24422c3b40791a4273f5f2819d5c0b818":{"kind":"REWORK","_number":3,"created":"2025-10-09 14:20:29.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/3"}}},"commit":{"parents":[{"commit":"39039716bff1d69aa938d7c0ae6f7e5eb69e5049","subject":"Add thread info logging to volume manager for debug purposes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39039716bff1d69aa938d7c0ae6f7e5eb69e5049"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-10-09 14:20:08.000000000","tz":0},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c63be9f24422c3b40791a4273f5f2819d5c0b818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c63be9f24422c3b40791a4273f5f2819d5c0b818"}]},"branch":"refs/heads/master"},"8917649207ac01fbfd559d0a52a33cba52db24c6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-09 18:31:56.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/4"}}},"commit":{"parents":[{"commit":"7db6e310882e339318d1acc98fd5fd82fc17666a","subject":"Add thread info logging to volume manager for debug purposes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7db6e310882e339318d1acc98fd5fd82fc17666a"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-10-09 18:30:38.000000000","tz":0},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8917649207ac01fbfd559d0a52a33cba52db24c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8917649207ac01fbfd559d0a52a33cba52db24c6"}]},"branch":"refs/heads/master"},"e3935c9713f2afe3db4c9fb376c8804c162ac090":{"kind":"REWORK","_number":5,"created":"2025-10-09 18:46:14.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/5"}}},"commit":{"parents":[{"commit":"276cff00b7a69527a4b4b2871cc6bd497387cf93","subject":"Add thread info logging to volume manager for debug purposes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/276cff00b7a69527a4b4b2871cc6bd497387cf93"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-10-09 18:44:58.000000000","tz":0},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3935c9713f2afe3db4c9fb376c8804c162ac090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3935c9713f2afe3db4c9fb376c8804c162ac090"}]},"branch":"refs/heads/master"},"c57624b28ecceb532d97f5c893b6bfd3973d375f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-11-19 17:02:44.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/6"}}},"commit":{"parents":[{"commit":"ce4a25dce90e9eedaf5a71f14e0354cee9908ab9","subject":"Add thread info logging to volume manager for debug purposes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce4a25dce90e9eedaf5a71f14e0354cee9908ab9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-11-19 17:02:44.000000000","tz":0},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c57624b28ecceb532d97f5c893b6bfd3973d375f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c57624b28ecceb532d97f5c893b6bfd3973d375f"}]},"branch":"refs/heads/master","description":"Rebase"},"452bfd1384ea1ecbdef833479652b0de86316b94":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-12-09 18:19:06.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/7"}}},"commit":{"parents":[{"commit":"c6efaf36cbcbc094899e0be70236497f9d3c28d9","subject":"Add thread info logging to volume manager for debug purposes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6efaf36cbcbc094899e0be70236497f9d3c28d9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-12-09 18:17:48.000000000","tz":-300},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/452bfd1384ea1ecbdef833479652b0de86316b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/452bfd1384ea1ecbdef833479652b0de86316b94"}]},"branch":"refs/heads/master"},"64204ae0a765724ba604da754e8a158b9f00c775":{"kind":"REWORK","_number":8,"created":"2025-12-09 18:27:15.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/8"}}},"commit":{"parents":[{"commit":"22d2fa2ce0c26966a358bd763f316e5aa469d49b","subject":"Tests: Add extend_volume to fake driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22d2fa2ce0c26966a358bd763f316e5aa469d49b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-12-09 18:26:01.000000000","tz":0},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64204ae0a765724ba604da754e8a158b9f00c775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64204ae0a765724ba604da754e8a158b9f00c775"}]},"branch":"refs/heads/master"},"da313326f64e6709696bbf9fe61734fb463c012d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-02-18 15:23:24.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/9"}}},"commit":{"parents":[{"commit":"93460c4503cbefeedc7a9aa0e4187855fadb8218","subject":"Tests: Add extend_volume to fake driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/93460c4503cbefeedc7a9aa0e4187855fadb8218"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-02-18 15:23:24.000000000","tz":0},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da313326f64e6709696bbf9fe61734fb463c012d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da313326f64e6709696bbf9fe61734fb463c012d"}]},"branch":"refs/heads/master","description":"Rebase"},"280994f9b655712f344d98c235c6827d25a6625e":{"kind":"REWORK","_number":10,"created":"2026-02-19 15:38:47.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/10"}}},"commit":{"parents":[{"commit":"16c6d302db71c6431bc8853a9a9ff43f38865a84","subject":"WIP: cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16c6d302db71c6431bc8853a9a9ff43f38865a84"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-02-19 15:31:17.000000000","tz":-300},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/280994f9b655712f344d98c235c6827d25a6625e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/280994f9b655712f344d98c235c6827d25a6625e"}]},"branch":"refs/heads/master"},"a997568a46e319c67777525249d3dd96ea4e54c8":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-02-23 17:49:48.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/11"}}},"commit":{"parents":[{"commit":"de57968c69a7de7fc53ccc2e2111ae1c6cfdb5f0","subject":"WIP: cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de57968c69a7de7fc53ccc2e2111ae1c6cfdb5f0"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-02-23 17:48:32.000000000","tz":-300},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a997568a46e319c67777525249d3dd96ea4e54c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a997568a46e319c67777525249d3dd96ea4e54c8"}]},"branch":"refs/heads/master"},"19c180cc919ea5131352f0e4001ff2157145633c":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-02-27 20:02:52.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/88/959388/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/12"}}},"commit":{"parents":[{"commit":"6e8912bfcb1be1ab72ded9ac0e5f82f0f0d0d323","subject":"WIP: cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e8912bfcb1be1ab72ded9ac0e5f82f0f0d0d323"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-27 19:25:21.000000000","tz":-180},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19c180cc919ea5131352f0e4001ff2157145633c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19c180cc919ea5131352f0e4001ff2157145633c"}]},"branch":"refs/heads/master"},"a60114d04a67d000c0dfe9079b5bb3ded86633a8":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-03-05 20:41:41.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/13"}}},"commit":{"parents":[{"commit":"9744e71bf44fb154b22c6a9e563c46b22d0000f7","subject":"RBD changes for eventlet removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9744e71bf44fb154b22c6a9e563c46b22d0000f7"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-05 20:07:36.000000000","tz":0},"subject":"WIP cinder-scheduler threading prototyping","message":"WIP cinder-scheduler threading prototyping\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a60114d04a67d000c0dfe9079b5bb3ded86633a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a60114d04a67d000c0dfe9079b5bb3ded86633a8"}]},"branch":"refs/heads/master"},"e0092d6e5757eeafaab8632637873988f4be0901":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":14,"created":"2026-03-10 17:49:40.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/14"}}},"commit":{"parents":[{"commit":"d5f967dd35aea2ee7040b1ce2666a9c9374028b1","subject":"Merge \"NetApp ONTAP MirrorAllSnapshots snapmirror replication failing issue fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5f967dd35aea2ee7040b1ce2666a9c9374028b1"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-10 17:48:44.000000000","tz":-240},"subject":"Remove eventlet.sleep calls from scheduler","message":"Remove eventlet.sleep calls from scheduler\n\nReplace these with time.sleep().\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0092d6e5757eeafaab8632637873988f4be0901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0092d6e5757eeafaab8632637873988f4be0901"}]},"branch":"refs/heads/master"},"caaeae085af95af2dd3ae07cd62a02b2acc8e638":{"kind":"REWORK","_number":15,"created":"2026-03-11 12:26:05.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/88/959388/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/959388/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/959388/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/959388/15"}}},"commit":{"parents":[{"commit":"d5f967dd35aea2ee7040b1ce2666a9c9374028b1","subject":"Merge \"NetApp ONTAP MirrorAllSnapshots snapmirror replication failing issue fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5f967dd35aea2ee7040b1ce2666a9c9374028b1"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 20:54:32.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-11 12:25:58.000000000","tz":-240},"subject":"Remove eventlet.sleep calls from scheduler","message":"Remove eventlet.sleep calls from scheduler\n\nReplace these with time.sleep().\n\nChange-Id: I12838d93ebd3328e163c7f9a32eeaeb03c42f1c7\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/caaeae085af95af2dd3ae07cd62a02b2acc8e638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/caaeae085af95af2dd3ae07cd62a02b2acc8e638"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
