)]}'
{"id":"openstack%2Fcinder~959886","triplet_id":"openstack%2Fcinder~master~I82e7a4edd24c27baed28b9b1eef7c09ce6e46e9d","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"37923":{"account":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"last_update":"2025-09-05 19:39:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I82e7a4edd24c27baed28b9b1eef7c09ce6e46e9d","subject":"Hitachi: add support for Hitachi VSP One B20","status":"ABANDONED","created":"2025-09-05 17:10:26.000000000","updated":"2025-09-06 05:43:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a664b1c16eb3e3290b0d191646d76f172c7e605c","_number":959886,"virtual_id_number":959886,"owner":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-05 19:09:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-05 17:14:06.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-05 17:19:17.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-09-05 18:01:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-05 18:58:58.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-09-05 19:09:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-06 05:43:11.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"7d30772fd43c1a3309f362ac7492ed63401e9175","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-09-05 17:10:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5877905944469e73541ae4f75827beab6b23d46","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-05 17:14:06.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"3a67dbb8be31c7ab98c984247df493d822c6ed1f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-05 17:19:17.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/476/","accounts_in_message":[],"_revision_number":1},{"id":"b4926a90d2517b73fd6bc47df282ea00211b8a23","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-05 17:41:26.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959886/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/959886/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"f02cf87cc61195bab49fb0d55bafa7384fefc467","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 18:01:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caeed5bde3ed415e8772c202abb6e4f1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3023b05e5b62445eb996beca0ce6867d : SUCCESS in 19m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/82e5289f1bc54e44b8d6a9196b591519 : SUCCESS in 23m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d515ce6c7a544843869bba8bb0b9ae33 : SUCCESS in 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ebb36693707dc61e26fba20143cfb105f4257db","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-05 18:58:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1699/.","accounts_in_message":[],"_revision_number":1},{"id":"539eec8c29584f4fb574b72dac9f703f38b7093e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-05 19:05:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1835/.","accounts_in_message":[],"_revision_number":1},{"id":"8a4364f225b8372c01fc1ea9aa012c0e7b3cf62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 19:09:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7b5b2a1f7fc431f8d69e4168f16bf24\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8b531f40e2b249b98e3162fb386dd233 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ed10c3c0f674a078c658452eabfed2b : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23259cc30a274622aa5d22e0baabdc36 : SUCCESS in 11m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ede1242c5d9d42d0839c872964504b66 : SUCCESS in 8m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37c1bda69c624d3b8bda760e3e5582b6 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fda3efa96614994baf2b49c9e627f7d : SUCCESS in 7m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/411cabfcc7214f028701145ec2ceb5ed : SUCCESS in 28m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0557219574264a87a284374867875bd3 : SUCCESS in 1h 01m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eafb8864bd894affb92af27b74a8daea : SUCCESS in 1h 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d96f24a9e4344f79be8e6a166c104114 : SUCCESS in 33m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/03bba11f5fe145a6866b370f1efb3c9c : FAILURE in 23m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8645fec684694ef0a2aa6556a6d1c9e1 : SUCCESS in 3m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/138c694dbd684151b909fb990528ab9b : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2577eca52db24eafa54c330e010a23f6 : SUCCESS in 5m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5343fcfbbdc84c27b13a960ed3eb7e7d : SUCCESS in 5m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13b82250e2e34a07974271db31534c90 : FAILURE in 26m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d14e4a74b0e84b4a9ef6c2e3aa8ed089 : SUCCESS in 3m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b76c3df22f434a2aa6b7f19c96cfaca1 : SUCCESS in 1h 14m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6054fe60348c4986810fcba944c7ac91 : FAILURE in 1h 42m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00a1da4b0ed94e9084e5624eb9677fe3 : SUCCESS in 1h 52m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/776efcd352f248158eaca9552f8c7c89 : FAILURE in 10m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4cc8fc512796489cbd4d6cbd4c179a62 : SUCCESS in 33m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/28d5aac69d8b42b1b5cb6c553a3b9966 : SUCCESS in 36m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f2428636499a4be289faf8deab71c321 : SUCCESS in 1h 38m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/914351fbff0146f39e3e5213be2382a1 : SUCCESS in 23m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/40eb8ac6cc7b476a92de465f84406c10 : SUCCESS in 1h 45m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1fc08b5c975b4df8b5aaaf35818a9dff : FAILURE in 11m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1236423dee5445c38bb13f954c988733 : SUCCESS in 48m 54s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8215fc9d46fa4a548923bc548f7c35f8 : SUCCESS in 1h 03m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01f226506d9a49cbb1d1082ab312c594 : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d5a3bf3336406ad596653e49d9f156097abcaa25","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-09-05 19:39:45.000000000","message":"Abandoned\n\nerror creating","accounts_in_message":[],"_revision_number":1},{"id":"a664b1c16eb3e3290b0d191646d76f172c7e605c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-06 05:43:11.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS,\\nlogs at https://oscilogs.quobyte.com/logs/${JOB_ID} .","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"417b33a4d1ad00a2672a6902bd3cd89b9166104a","revisions":{"417b33a4d1ad00a2672a6902bd3cd89b9166104a":{"kind":"REWORK","_number":1,"created":"2025-09-05 17:10:26.000000000","uploader":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"ref":"refs/changes/86/959886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/86/959886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/959886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/959886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/959886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/86/959886/1"}}},"commit":{"parents":[{"commit":"2a409a609b71aa4ec22dc90753af111283f4f750","subject":"Merge \"[IBM SVf driver] Adding support for temporary volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a409a609b71aa4ec22dc90753af111283f4f750"}]}],"author":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-05-28 20:44:51.000000000","tz":0},"committer":{"name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-09-05 17:10:10.000000000","tz":-420},"subject":"Hitachi: add support for Hitachi VSP One B20","message":"Hitachi: add support for Hitachi VSP One B20\n\nChange-Id: I82e7a4edd24c27baed28b9b1eef7c09ce6e46e9d\nSigned-off-by: Anthony Gamboa \u003canthony.gamboa@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/417b33a4d1ad00a2672a6902bd3cd89b9166104a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/417b33a4d1ad00a2672a6902bd3cd89b9166104a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
