)]}'
{"id":"openstack%2Fcinder~960617","triplet_id":"openstack%2Fcinder~master~Id3b9ed1056d178314d45f536bc9b60a3091ae729","project":"openstack/cinder","branch":"master","attention_set":{"38280":{"account":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"last_update":"2025-09-11 18:56:14.000000000","reason":"Someone else replied on the change"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-01-05 06:37:28.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2025-11-08 10:58:23.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2025-09-11 15:44:32.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Id3b9ed1056d178314d45f536bc9b60a3091ae729","subject":"NetApp: Add retype support for front-end QoS changes","status":"NEW","created":"2025-09-11 15:05:51.000000000","updated":"2026-01-05 13:55:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"a9749da12db038ff13cb95b01a61f4d8e33f5d0e","_number":960617,"virtual_id_number":960617,"owner":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-08 14:12:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"all":[{"value":-1,"date":"2025-12-17 12:16:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":-1,"date":"2026-01-05 07:11:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":-1,"date":"2026-01-05 06:37:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-11 15:06:01.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-09-11 15:07:18.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-11 17:00:51.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-11 17:06:57.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-11 17:11:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-11 17:13:50.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-09-11 17:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-11 18:16:14.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-11 18:56:14.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-10-31 16:18:16.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2025-11-10 08:21:31.000000000","updated_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"reviewer":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"state":"CC"},{"updated":"2025-12-11 08:59:10.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"CC"},{"updated":"2025-12-11 08:59:10.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"CC"},{"updated":"2025-12-17 11:35:51.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-12-17 12:16:49.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-01-05 06:37:28.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-01-05 07:11:12.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"}],"messages":[{"id":"8df0febe4ffbf2b084ee1c0adbd8f34ef176d3b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"date":"2025-09-11 15:05:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4b01d703813781e7849a9179737923aa0fbc1c6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 15:06:01.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/557/","accounts_in_message":[],"_revision_number":1},{"id":"0f5498be9f2a1c55982c59e50c329fb6b7c08cfe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-11 15:07:18.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"336207873de0b357da0a878ca58f7bf733b797f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"date":"2025-09-11 15:10:15.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"433910e3b84bae15015eae1817149fc9550a62e8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-11 15:11:42.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"fd30fc46203ae5e6b099ad8b626cb9c2c86962e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"date":"2025-09-11 15:14:03.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1cb30f9d3b09b845a9325c550fd08e156c0abb4a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-11 15:15:29.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"b6bd2809724436752ef888999244c13628db14e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 15:24:56.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/960617/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/960617/1/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"48dc174e9463fc76d5f50353fbb0eb844c5f4290","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 15:24:57.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/558/","accounts_in_message":[],"_revision_number":2},{"id":"42d2e6c90419509b801a424823fb850f9dbcb19f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 15:44:32.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/960617/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/960617/2/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"a4e3d6ec96ee802dc6109ee61c73699c39d92e05","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 15:44:33.000000000","message":"Patch Set 3:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/559/","accounts_in_message":[],"_revision_number":3},{"id":"b532f60f82da6aea6f10207a05070cc7b40e48b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 16:04:05.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/960617/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/960617/3/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"9a5bd18c4ac3b2d712a3dbeea451f0644519ad7a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-11 17:00:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3d3922b6ea724dffb93acfe11f20d8fc : SUCCESS in 49m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/43949ef7b63a4b2ab8accd51be8fa1c0 : FAILURE in 58m 37s","accounts_in_message":[],"_revision_number":3},{"id":"9598d44f9c6382b660b542553dff13d4c8070a0e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-11 17:06:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/Id3b9ed1056d178314d45f536bc9b60a3091ae729 .","accounts_in_message":[],"_revision_number":1},{"id":"2596afeceb8b4367d0202ce057491513a0842265","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 17:11:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f60d0fc186cc4609886abb3acbf427ae\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1cc4ca4a233b46c786c8d1b5a8d2aa49 : SUCCESS in 20m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c1e65e01df7b4caf8c739c71229f20b4 : SUCCESS in 22m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b9f03d0387c440c8b1041aaebb909c8b : SUCCESS in 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4a84e8a8b0ea1922abe226cb4153451ae1972576","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-11 17:13:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/960617 : FAILURE in 1h 17m 43s","accounts_in_message":[],"_revision_number":3},{"id":"a2ce0507e61bb41bcc88b96b75035e11cc35c1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 17:50:48.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca9f17a2dc944ab08768e396af57e861\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6f35935e59a4f08bf0c1e64f9290715 : FAILURE in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2562510704f54d3bbfa5d6134b3df420 : SUCCESS in 7m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd79220be64b4455a4fbd9b402704714 : POST_FAILURE in 12m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dea2377135c84da2a077e9de85e02e20 : SUCCESS in 13m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21fcb0ed2964440490483eb3c9f9e16d : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/b47c1266fefa4dbaab15336dab971173 : SUCCESS in 30m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/10f190fc061e40c895058518d1c3a982 : SUCCESS in 1h 06m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb1c2b99714a40cfb55336beb4ebbe7e : SUCCESS in 1h 31m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fbf15cf7f944564baf68d3c396422b1 : SUCCESS in 32m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/038a7df1b99145d9b72d6448540d0034 : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/29aa16f0611246e2aa61d44514a48f61 : SUCCESS in 6m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a1e98dd81354c98a285ea40cb62d37b : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b2d0e72b00e4b429379c89440459e9b : SUCCESS in 5m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/12dace626cbe415d8e6b02cd3cb4fb54 : SUCCESS in 8m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/94091919d9ed450aad7a80aa90b34c74 : FAILURE in 30m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7af0802e964c44bf83b5fcc1e8489eca : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/83c191d6aa6c4e9bb0515438940d0c85 : SUCCESS in 1h 53m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/21d4b3e978f04a87b5a743fdb128327c : FAILURE in 2h 34m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d63dde2e418c46eea83f6ba1484631f2 : SUCCESS in 1h 48m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fea61cba9a484d6d98f0a026670a889c : FAILURE in 19m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ee84880bb68d49d08aa680f122a6bc2f : SUCCESS in 30m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e2437cb123b948dcbb9771b10e185665 : SUCCESS in 34m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c758dffafa494d0b839cc5bd260e81ac : SUCCESS in 1h 30m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/acb8dea763aa4f36b861a7238c7ef1aa : SUCCESS in 23m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fbed61c406cd4369b42f0ed1c3d7423f : SUCCESS in 1h 33m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/548abe599f7444d28f81e1e38fd07ff1 : FAILURE in 20m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54aef3f039594ef9bccdb5ced3f8213e : SUCCESS in 1h 23m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32ad0c2dd0974772bc5539276f8362ba : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":3},{"id":"3b6073dbcf41ec838ad7748a2095a28b5f94b629","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-11 18:16:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/960617 : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":3},{"id":"938b39d65a50e5defdb111c409ef504f74dbb161","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-11 18:28:33.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/Id3b9ed1056d178314d45f536bc9b60a3091ae729 .","accounts_in_message":[],"_revision_number":2},{"id":"4a68fbf04d433622d742119c2f71194125f74758","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 18:56:14.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1902/.","accounts_in_message":[],"_revision_number":1},{"id":"e2fcd715b137d3958542dd7d359d2edb8c1117bb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-11 18:58:19.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/Id3b9ed1056d178314d45f536bc9b60a3091ae729 .","accounts_in_message":[],"_revision_number":3},{"id":"b9fd5740dc889b4816afc294cd537bf3862fd175","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 19:43:19.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1768/.","accounts_in_message":[],"_revision_number":1},{"id":"deaa5cd5452029f6d370f0b68484cea580b0e347","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 20:12:59.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1903/.","accounts_in_message":[],"_revision_number":2},{"id":"05797c20f9edcffdbe11a3de7e20c7d7c7727538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"date":"2025-09-11 20:42:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4174a901e7aff28b69c5a9dc1f6db021613149e6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-11 20:43:48.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"453a759c299105e6f3573b348ff0ca536a1274b3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 20:44:30.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1769/.","accounts_in_message":[],"_revision_number":2},{"id":"e77a4895a4484b8ccaedcc0ea059db05caa3382d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 20:46:41.000000000","message":"Patch Set 4:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/566/","accounts_in_message":[],"_revision_number":4},{"id":"f6eed0816301b6cd29b4b72ea27829f14633231b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-11 21:05:14.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/960617/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/960617/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"4e219e30b3e318264a284a689025633643667fa9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 21:05:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a9b4cb9ae7b46388aed52ef27946b16\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b66bcf91f72e4593924647cddc3fe292 : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/28c1ad3d6c364641b42a51b3ae4f7f59 : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f1f995b6157e45879b44cfe3f59de93a : SUCCESS in 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ffb5e6e0ba3c2eccdd75153f312c46bc96b0329f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 21:16:43.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1904/.","accounts_in_message":[],"_revision_number":3},{"id":"022239c93f150d0bb86126a872cfbc07eb1fe2ac","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-11 21:50:08.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1770/.","accounts_in_message":[],"_revision_number":3},{"id":"e2ea88875d920f4d5e172b46cda0595dab1773c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 22:40:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe2ee1a2aacb466da5672d50d98fe690\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74bfcc9d6fbb4cdf87610634f9183b90 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cdc0ec682c7443db823e3f1e177e3c5 : SUCCESS in 7m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14475b85a39c43c9a7be36ca19eb4fd8 : SUCCESS in 11m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92edfb11c8524b6fb2da7aff98f0e073 : SUCCESS in 14m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c1d5c5e3f4a4f1ca553f78914d9f3bc : SUCCESS in 16m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/ea39470c0f4f422db61159a66542dcf2 : SUCCESS in 39m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf6f2fa8ab7249e48d14eedf0ecef6e4 : SUCCESS in 43m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fdda09369b1b4bdca4a9dc46d4c0bf7d : SUCCESS in 1h 46m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afce38ce1e4749beb9be4c0e2b0bcfe8 : SUCCESS in 48m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/73c50bf4d3354e8c93b198a5e547018a : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c6dc146a636b4a2f9f53833230067585 : SUCCESS in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/10468604368c4a6b83a33ea90a8db57a : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c591109a77f24f5db499f28c12134b24 : SUCCESS in 6m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/224b3e1d01cb42909d626b796bfb013e : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e80566ce2c054161a8730c12a2d8d1f0 : FAILURE in 28m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8dd7fcce01ac4ccdb89735d84fc6afe3 : SUCCESS in 4m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/84d3383398914fefa2aea103f4922ca4 : SUCCESS in 1h 53m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1c914f7e5d8544428d9e48442f732f84 : FAILURE in 1h 34m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3467495571c24548952df16a46cebe7d : SUCCESS in 1h 18m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/68b0cf417d8346b887939c7bc82dd427 : FAILURE in 8m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cdc9a039a64a453b821a48be1bd72f37 : SUCCESS in 31m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7403330cf5a84ab591ac481bfa1b1b99 : SUCCESS in 58m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/157d9a9fb87b4c0fad43ee010267950b : SUCCESS in 1h 30m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/43d15b5e25254edbaf4b8ac3a638cff1 : SUCCESS in 44m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/413add180cf648eea7a5fedcac465bdf : SUCCESS in 1h 41m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/57512e88ad4340c6a54fa609affd7e25 : FAILURE in 16m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75f6c83fe50747d68e4c2da677dbdcd5 : SUCCESS in 1h 48m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af4440df6c014e30abc703991adab757 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":4},{"id":"2d5279d5a9174d1e13cbb791ae5be973c7c78a1e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-11 23:12:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/30572331a130430e996427d60fe26434 : SUCCESS in 48m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e8e9317c1dac44df8b7428d9caf7daeb : FAILURE in 52m 32s","accounts_in_message":[],"_revision_number":4},{"id":"31c0890575e9fe2e60c9041b6842855ff82dddbe","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-12 01:02:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/960617 : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":4},{"id":"aea66d0ba7126a24c17292687e20d05f69bb240f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-12 01:15:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/960617 : FAILURE in 1h 17m 15s","accounts_in_message":[],"_revision_number":4},{"id":"ee14eca8f03d05ddfdfeea7b7990831f38128051","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-12 02:10:29.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE,nlogs at https://oscilogs.quobyte.com/logs/Id3b9ed1056d178314d45f536bc9b60a3091ae729 .","accounts_in_message":[],"_revision_number":4},{"id":"bcc6e4d8870564f87253d9b29fb2018c3d0d61e2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-12 04:44:09.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1911/.","accounts_in_message":[],"_revision_number":4},{"id":"6fd8e7d01df7923fe81cd358f65cb4e24db58078","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-12 06:50:07.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1777/.","accounts_in_message":[],"_revision_number":4},{"id":"43779daa8ec11f7051b67e5762719ba09e83aba4","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-10-31 16:18:16.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"792cea49f73c6752d5615557439cdd31f07c3a5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"date":"2025-11-08 10:58:23.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"76672722b9d0ff29326a3173b8e6a267a2be3f44","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-08 10:59:18.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"8ae32360487620ed9c795dc1a324ca9270554dc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-08 11:23:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3df040e083c54b2182950c45836081b1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/851889787d8d4c439c7386b05b4283e3 : SUCCESS in 17m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/004ba5b8fff44e4da203a68ee2f6f7fd : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/de78af3b7e584b7291191ec1c429d870 : SUCCESS in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"72a1ca5217c4cc874fde4355337963c29595ef6e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-08 11:56:54.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id3b9ed1056d178314d45f536bc9b60a3091ae729 .","accounts_in_message":[],"_revision_number":5},{"id":"79be0a24123e10a242fbcc02bb6715f9352f63c7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-08 12:14:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/37ed5596e66148108ce1bcd128030690 : SUCCESS in 50m 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fabc75459b55464c84e083b03264267f : FAILURE in 1h 03m 19s","accounts_in_message":[],"_revision_number":5},{"id":"5cb155d26fcbec873cdc972e659feead946b60cb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-08 12:36:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2347/.","accounts_in_message":[],"_revision_number":5},{"id":"8d879522e801f5975086a7a6b624e1079d1fef57","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-08 12:39:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2461/.","accounts_in_message":[],"_revision_number":5},{"id":"16472424640d4401b8767dfcae8e5134a8e902fb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-08 12:40:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/960617 : FAILURE in 1h 12m 16s","accounts_in_message":[],"_revision_number":5},{"id":"5893b6012ec536318a7a5dd8ebd2bf883b7f3c0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-08 14:12:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaba77a42883423990c26f8492f25194\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff8bf44104454a999df6115ef4807679 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ec654fdc19742db941105c1362f1ee4 : SUCCESS in 11m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21adfcadf00748cca4838babb7340db1 : SUCCESS in 15m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd9fb191d494f28b63b9bd3d3e0497e : SUCCESS in 15m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/1f713e40b2604a398ea214c7d8b8d255 : SUCCESS in 1h 00m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3515da7b6034710b8530a0844b1243c : SUCCESS in 58m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b10561716024aa1b5b0c549438bd014 : SUCCESS in 1h 19m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e185deb5f6144c5b42e12c537811a61 : SUCCESS in 1h 05m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/005c2e0f530346ccb51a1e753aac9d20 : SUCCESS in 13m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28c984e8bf0b4863919d773851a3eaab : SUCCESS in 21m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/73de0294247b43ba955ce4a1a1cb8049 : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a7187c2225184865902467e1905e1a79 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3dc801d492ab4400a91796d1c5076686 : SUCCESS in 5m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/afd53297cf3343e2b7517ce23cf3d706 : SUCCESS in 6m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4ee6c642cefa412ba6fbfbbf0d70c281 : FAILURE in 31m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7286fb68da846dd8399be8bd3568eaf : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a2b3a0fdbcef44adabc6d9051e1e1005 : SUCCESS in 1h 38m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3751658c1d784329925182d978ac3ec8 : TIMED_OUT in 3h 05m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bbca8ab743ac4b1ca2792836bcee156a : SUCCESS in 1h 58m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a6dccca1d77c4251a750c5fd74898663 : SUCCESS in 2h 00m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4486e20457c0469091eb3236de9530f1 : SUCCESS in 36m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/80a81a9111144b3e92184a8dcb7c73c4 : SUCCESS in 1h 17m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5abd86721db0498395a9773f6ce651ba : SUCCESS in 1h 39m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1382521167bf4560b29cbd722406613b : SUCCESS in 46m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/33e2c8858436499e968c92acff5111d4 : SUCCESS in 2h 04m 21s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3d0f075369734d8790c7a987d52b3ef5 : TIMED_OUT in 3h 09m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/84f8ae86d86a4995b45ce071912d07e1 : SUCCESS in 1h 23m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/707c957bea2d4d238d1e739f73ddb284 : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":5},{"id":"45bea6eb88161e31175088b2365852e73b6b8515","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-11-10 08:21:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5fa6cb83f8354a69b262cb53579f0c1f1ce9902e","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2025-12-17 10:45:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7d7a29bd2efb51e177c006abf90c3781b08d365a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-12-17 11:35:51.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/960617/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/8bfca0a/ : SUCCESS in 49m 08s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/960617/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/681b255/ : SUCCESS in 38m 07s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"785da1c75ede21a9b7d8fbc999e44e44ac13c4a7","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2025-12-17 12:16:49.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bfeeaaaed4d1495945dcec40bd9ac1a564f1d0cd","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-01-05 06:37:28.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"47cd2f1261af11c625947a7f61680ab6c0c3b80b","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-05 07:11:12.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a9749da12db038ff13cb95b01a61f4d8e33f5d0e","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-05 13:55:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"67facc79f39eae11089f77b0768edf1adc0c7d00","revisions":{"4af88a9a5ad06e0dde1016db55a8ace210bc0082":{"kind":"REWORK","_number":1,"created":"2025-09-11 15:05:51.000000000","uploader":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"ref":"refs/changes/17/960617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/960617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/960617/1"}}},"commit":{"parents":[{"commit":"fe5861dc19db2b67845b83f50d8fd56a9b978ddf","subject":"Merge \"[Pure Storage] Consistent hostnames across replicated backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe5861dc19db2b67845b83f50d8fd56a9b978ddf"}]}],"author":{"name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","date":"2025-09-11 13:36:41.000000000","tz":120},"committer":{"name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","date":"2025-09-11 15:05:33.000000000","tz":120},"subject":"NetApp: Add retype support for front-end QoS changes","message":"NetApp: Add retype support for front-end QoS changes\n\nEnable in-place volume retype when only front-end QoS specs change,\navoiding costly migration operations. The retype method returns True\nfor front-end QoS changes that can be handled by Nova/libvirt, and\nFalse for other changes that require volume migration.\n\nThis improvement allows users to modify front-end QoS specifications\nwithout triggering volume migration, providing better performance\nand user experience.\n\nImplements: retype method for NetApp block storage drivers\nChange-Id: Id3b9ed1056d178314d45f536bc9b60a3091ae729\nSigned-off-by: Ali Yousefian \u003caliyousefiaan@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4af88a9a5ad06e0dde1016db55a8ace210bc0082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4af88a9a5ad06e0dde1016db55a8ace210bc0082"}]},"branch":"refs/heads/master"},"8d9ae4898aed09e6bffd5013183388bb48bce2cb":{"kind":"NO_CHANGE","_number":2,"created":"2025-09-11 15:10:15.000000000","uploader":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"ref":"refs/changes/17/960617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/960617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/960617/2"}}},"commit":{"parents":[{"commit":"fe5861dc19db2b67845b83f50d8fd56a9b978ddf","subject":"Merge \"[Pure Storage] Consistent hostnames across replicated backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe5861dc19db2b67845b83f50d8fd56a9b978ddf"}]}],"author":{"name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","date":"2025-09-11 13:36:41.000000000","tz":120},"committer":{"name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","date":"2025-09-11 15:10:02.000000000","tz":120},"subject":"NetApp: Add retype support for front-end QoS changes","message":"NetApp: Add retype support for front-end QoS changes\n\nEnable in-place volume retype when only front-end QoS specs change,\navoiding costly migration operations. The retype method returns True\nfor front-end QoS changes that can be handled by Nova/libvirt, and\nFalse for other changes that require volume migration.\n\nThis improvement allows users to modify front-end QoS specifications\nwithout triggering volume migration, providing better performance\nand user experience.\n\nImplements: retype method for NetApp block storage drivers\nChange-Id: Id3b9ed1056d178314d45f536bc9b60a3091ae729\nSigned-off-by: Ali Yousefian \u003caliyousefiaan@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d9ae4898aed09e6bffd5013183388bb48bce2cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d9ae4898aed09e6bffd5013183388bb48bce2cb"}]},"branch":"refs/heads/master"},"24d7fc020b91017f4380d5c9f97a9f6c03de4a80":{"kind":"NO_CHANGE","_number":3,"created":"2025-09-11 15:14:03.000000000","uploader":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"ref":"refs/changes/17/960617/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/960617/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/960617/3"}}},"commit":{"parents":[{"commit":"fe5861dc19db2b67845b83f50d8fd56a9b978ddf","subject":"Merge \"[Pure Storage] Consistent hostnames across replicated backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe5861dc19db2b67845b83f50d8fd56a9b978ddf"}]}],"author":{"name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","date":"2025-09-11 13:36:41.000000000","tz":120},"committer":{"name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","date":"2025-09-11 15:13:49.000000000","tz":120},"subject":"NetApp: Add retype support for front-end QoS changes","message":"NetApp: Add retype support for front-end QoS changes\n\nEnable in-place volume retype when only front-end QoS specs change,\navoiding costly migration operations. The retype method returns True\nfor front-end QoS changes that can be handled by Nova/libvirt, and\nFalse for other changes that require volume migration.\n\nThis improvement allows users to modify front-end QoS specifications\nwithout triggering volume migration, providing better performance\nand user experience.\n\nImplements: retype method for NetApp block storage drivers\nChange-Id: Id3b9ed1056d178314d45f536bc9b60a3091ae729\nSigned-off-by: Ali Yousefian \u003caliyousefiaan@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24d7fc020b91017f4380d5c9f97a9f6c03de4a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24d7fc020b91017f4380d5c9f97a9f6c03de4a80"}]},"branch":"refs/heads/master"},"88a0f06ffcc492adda5460b0ce6a7b74c4faca39":{"kind":"REWORK","_number":4,"created":"2025-09-11 20:42:21.000000000","uploader":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"ref":"refs/changes/17/960617/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/960617/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/960617/4"}}},"commit":{"parents":[{"commit":"fe5861dc19db2b67845b83f50d8fd56a9b978ddf","subject":"Merge \"[Pure Storage] Consistent hostnames across replicated backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe5861dc19db2b67845b83f50d8fd56a9b978ddf"}]}],"author":{"name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","date":"2025-09-11 13:36:41.000000000","tz":120},"committer":{"name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","date":"2025-09-11 20:29:53.000000000","tz":120},"subject":"NetApp: Add retype support for front-end QoS changes","message":"NetApp: Add retype support for front-end QoS changes\n\nEnable in-place volume retype when only front-end QoS specs change,\navoiding costly migration operations. The retype method returns True\nfor front-end QoS changes that can be handled by Nova/libvirt, and\nFalse for other changes that require volume migration.\n\nThis improvement allows users to modify front-end QoS specifications\nwithout triggering volume migration, providing better performance\nand user experience.\n\nImplements: retype method for NetApp block storage drivers\nChange-Id: Id3b9ed1056d178314d45f536bc9b60a3091ae729\nSigned-off-by: Ali Yousefian \u003caliyousefiaan@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88a0f06ffcc492adda5460b0ce6a7b74c4faca39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88a0f06ffcc492adda5460b0ce6a7b74c4faca39"}]},"branch":"refs/heads/master"},"67facc79f39eae11089f77b0768edf1adc0c7d00":{"kind":"REWORK","_number":5,"created":"2025-11-08 10:58:23.000000000","uploader":{"_account_id":38280,"name":"Ali Yousefian","display_name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","username":"aliyousefiaan"},"ref":"refs/changes/17/960617/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/960617/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/960617/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/960617/5"}}},"commit":{"parents":[{"commit":"fe5861dc19db2b67845b83f50d8fd56a9b978ddf","subject":"Merge \"[Pure Storage] Consistent hostnames across replicated backends\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe5861dc19db2b67845b83f50d8fd56a9b978ddf"}]}],"author":{"name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","date":"2025-09-11 13:36:41.000000000","tz":120},"committer":{"name":"Ali Yousefian","email":"aliyousefiaan@gmail.com","date":"2025-11-08 10:58:21.000000000","tz":60},"subject":"NetApp: Add retype support for front-end QoS changes","message":"NetApp: Add retype support for front-end QoS changes\n\nEnable in-place volume retype when only front-end QoS specs change,\navoiding costly migration operations. The retype method returns True\nfor front-end QoS changes that can be handled by Nova/libvirt, and\nFalse for other changes that require volume migration.\n\nThis improvement allows users to modify front-end QoS specifications\nwithout triggering volume migration, providing better performance\nand user experience.\n\nImplements: retype method for NetApp block storage drivers\nChange-Id: Id3b9ed1056d178314d45f536bc9b60a3091ae729\nSigned-off-by: Ali Yousefian \u003caliyousefiaan@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67facc79f39eae11089f77b0768edf1adc0c7d00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67facc79f39eae11089f77b0768edf1adc0c7d00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
