)]}'
{"id":"openstack%2Fcinder~961431","triplet_id":"openstack%2Fcinder~stable%2F2025.1~I40263682af4735406d341e77ca90ee37cb361994","project":"openstack/cinder","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2025-11-10 19:24:22.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2025-11-10 19:24:22.000000000","reason":"Change was submitted"},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-11-10 19:24:22.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2025-11-10 19:24:22.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2025-11-10 19:24:22.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-11-10 19:24:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I40263682af4735406d341e77ca90ee37cb361994","subject":"[NetApp] Enabling total_volumes capability support","status":"MERGED","created":"2025-09-17 04:17:29.000000000","updated":"2025-11-10 19:25:40.000000000","submitted":"2025-11-10 19:24:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"961431","cherry_pick_of_change":959293,"cherry_pick_of_patch_set":11,"meta_rev_id":"1f86d457a072221ce0ddf4438e4b2bb5e4decc15","_number":961431,"virtual_id_number":961431,"owner":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-10 19:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-22 14:29:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":2,"date":"2025-09-22 13:33:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-22 14:29:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-17 04:17:55.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-09-17 04:20:16.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-09-17 04:20:16.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-09-17 04:20:16.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2025-09-17 04:20:16.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-09-17 04:20:16.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-09-17 04:38:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-17 04:55:44.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-17 04:56:00.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-09-17 05:23:53.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-09-17 06:10:45.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-17 07:04:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-17 08:33:40.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-09-22 13:33:39.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2025-09-26 21:04:58.000000000","updated_by":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"reviewer":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"state":"CC"},{"updated":"2025-10-22 14:29:43.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2025-10-22 21:24:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2025-11-05 07:01:57.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"2195b7de42011fb7819555d8126f84e324f61df1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-09-17 04:17:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8b206de3c7a6b12a77de6b25e7855362f4c2c23d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-17 04:17:55.000000000","message":"Patch Set 1:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/600/","accounts_in_message":[],"_revision_number":1},{"id":"bd133b2041ce638a83a29c6a90affa37a15b0c7a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-17 04:37:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/961431/1/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/961431/1/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"64f3f9dfbd097eed45cff6d4676af0697cebe430","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 04:38:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17c9d74749f04a94a6036455c2946c52\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42a306e2d1464c72b204f0ab43453b1b : SUCCESS in 18m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4dd70f4024664bce8dafe21bc20b6ac1 : SUCCESS in 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b3aa003f5f908d983b20bc1f827f770333bcdc6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-17 04:55:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/961431 : FAILURE in 36m 20s","accounts_in_message":[],"_revision_number":1},{"id":"02ce5a0cce72536e0b2f9e83023d931fc252a39b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-17 04:56:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/961431 : FAILURE in 36m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6d96d9de4bb1a451a304bd6ed645b015aa3ba364","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-17 05:23:53.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1945/.","accounts_in_message":[],"_revision_number":1},{"id":"1b69586839b9c3444e33443233db5146034948fd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-17 05:39:47.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1811/.","accounts_in_message":[],"_revision_number":1},{"id":"62aecad171d43a1a80f66123a0f3387345a82f99","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-17 06:10:45.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I40263682af4735406d341e77ca90ee37cb361994 .","accounts_in_message":[],"_revision_number":1},{"id":"6c4bc88cd68e4349992cf32012c2328153f0706d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 07:04:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9359b9a0271f4360905b14fccb3b285d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63736fcc9384f3dad8365a9886fc45d : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/354826a93f404bd2b781f724442499ef : SUCCESS in 12m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31f5c498e4f74481a5c086374880fc5c : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c98ec2756c4341cea8b7e17d48f575f9 : SUCCESS in 15m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/a48ca9e9d7464589a1c24e7ce66e2266 : SUCCESS in 56m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/91113b744f2d404f9c81378a8983e25e : SUCCESS in 57m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/41f4df4f6f594e1f8df26f1b9b414245 : SUCCESS in 1h 28m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c19cab054134681a34d919cfe3791a5 : SUCCESS in 1h 02m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/124afebc01154220afd417d2e4245e04 : SUCCESS in 14m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4251863942cd4dd6a8adfbea868e0224 : SUCCESS in 22m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ec41cbb084f74f9793d6cd684e52e3a7 : SUCCESS in 6m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4be5c9d8d37d4d8c913e0d80c4df6607 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0bb7011bc7da45a2ae614f219b5959ac : SUCCESS in 8m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/19409bca54bb4e53881345d4401df1a0 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/80f8278e406d4aea8b832e75ef936267 : FAILURE in 37m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/57ea61fccf224a038cac405599944609 : SUCCESS in 8m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/48695dbea60b4f02b71569343e491092 : SUCCESS in 2h 08m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/85896f7b27144a5e9bc6c379eb3db90e : SUCCESS in 2h 44m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bf8bf3e45c5046c99617eb5a8be62a11 : SUCCESS in 1h 35m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8e749ee25ce2447790484b78ecc05019 : SUCCESS in 1h 46m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c1a8b656fd4c4b1b809c0ac71c2b56a1 : SUCCESS in 34m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/50ba46e1db5f4927a6a5da38c7ac34d8 : FAILURE in 47m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5c719de4d4544633b36a246ca5b6a7f3 : SUCCESS in 1h 27m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9283f019450b4589b0240d766104ebd0 : SUCCESS in 41m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/874cb7064c2e40b4a245289458b0a9f3 : SUCCESS in 1h 30m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/704876a2fd604e67b2d394872b498580 : SUCCESS in 1h 31m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b43d8c728944086a8389186760a3aad : SUCCESS in 1h 58m 06s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6562a2f7bee84d95bb7c71969b4c14e0 : SUCCESS in 1h 24m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd80f3229e7149f5935f45c661033e6b : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a4b5bf406018bb2a345b738d0a75047ed07a9404","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-09-17 07:11:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2de5bcdb5280e8cae12d9a976a4306b583395a4e","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-09-17 07:40:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91ca8e7752b6f3fea2a6421f26d415aecd8122e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 07:41:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fafab9afdc2b4e49adce618b1a21a92d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9545d9a4e10246378e65b60dd8034916 : SUCCESS in 28m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d981d7b75b840818e8eb3290bfb29fc : SUCCESS in 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f9c4c27ab5f249b557092629932b5baf7c5ad54","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-17 08:33:40.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0edea254c62849e1b6a0869c29332266\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f0ed02f5ec1e4b629d26f52604ae005c : NODE_FAILURE Node request 200-0000078939 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b1fc4d8b4891478bb8f803267926c26d : NODE_FAILURE Node request 200-0000078940 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0ccc246425c34e4f909f1dddd5540d98 : SUCCESS in 1h 53m 33s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1795e6e958be4847971018a94adbdace : FAILURE in 1h 57m 57s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/db0ac14c5ad549e7a852e52699a40a2b : SUCCESS in 42m 42s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/149a571f640247c4bd4244a95be3ffb3 : SUCCESS in 1h 53m 34s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9445645a84314c939b4e44f087b1bcaf : FAILURE in 1h 53m 33s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/686dd73445064e86b6265f70f8fc0159 : SUCCESS in 1h 32m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7416d548f4d7f79835103d36c212eb0c8214a65c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 10:13:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28a0e07bb50341949c932c195be56e9c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f6e0f965e434f8899ffc6bc6b92c11b : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7663f6f654364ec68096ac7fcf73a693 : FAILURE in 12m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c22b147ae28c45749ed5e37fc5f76b5b : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62d1c70fabc3413db8657d8d5f105ef0 : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/1db329d2ed504b298286aa622177b5f2 : SUCCESS in 59m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab193073fde8423a98d0549ca7ac45a6 : SUCCESS in 44m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3b845475e76141c881a261f453e17931 : SUCCESS in 1h 26m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e33177fe8994b8098afdc7f5f044fc0 : FAILURE in 1h 07m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ce474db697e44fba1abc9c1d43a5f94 : SUCCESS in 14m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3f48f390f8fd4c1cb0e219b9a8c068ba : SUCCESS in 27m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/93d7de39d1b941549072afa92d9918c9 : SUCCESS in 6m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3114352e221046cd8a923afef235d5d8 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a280cae3ede0497aa56ee9b989bd51d7 : SUCCESS in 6m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bebd00a07f2745c5b8af992128b3a0eb : SUCCESS in 9m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cb8e65feaa404128b5741125e46ccfae : FAILURE in 36m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7e96b5d2a037474ab7072ab167137c96 : SUCCESS in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/711e51fe94a0491d836eb34cfa536bc8 : SUCCESS in 1h 57m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/04ed0c82c58b48739ed79088c86b33ee : SUCCESS in 2h 52m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fa8d5b56305846b4b44497ad2378db26 : SUCCESS in 1h 54m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1492bd3531594093afd029fadc1910b3 : SUCCESS in 1h 58m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/84be666e526a44cd9cf9492af6c5f9ec : SUCCESS in 38m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b1f9ce5c65034233a804c41ee51f0d09 : SUCCESS in 1h 09m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/40235772568540a984a37497bbe229e5 : SUCCESS in 1h 14m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b7f2366c89c24052bdaf6e1a6bec2158 : SUCCESS in 35m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/25867482db344803bea5130285b05d24 : SUCCESS in 1h 12m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0682a928ac2c43a59b225f34a53088ce : SUCCESS in 1h 18m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/542d65f7731443419554c42db653abde : SUCCESS in 1h 17m 56s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29e28d5b18624fb5a834d981d5addf6c : SUCCESS in 1h 11m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/443c443592ed41229bf1bcf7c3e407d0 : SUCCESS in 47m 39s","accounts_in_message":[],"_revision_number":1},{"id":"def3aa8f1e22a18ed2ce5357f80b0fdb5cf2ac40","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-09-17 17:05:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b208ab2a3cf7b6bed98cc448de096e77c62bc98b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 19:47:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9707eef18c124fa4ba1f6609548f4772\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b7710cc42cf478ea2735f60f3f80121 : SUCCESS in 16m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dda90abef7f94a1492119df8a5150109 : SUCCESS in 23m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04ce08bd0fcef2934121dec95dd709b4b123ffd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 21:59:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cd99ea776034bce945b7e9868f21714\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/066d1e13981e4d0599881de34a36e419 : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62eb4a0e8e7949caa2b13e4897640450 : SUCCESS in 13m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18672ff1de154aa680e9ee9e98c7091c : SUCCESS in 11m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364762c8494e4025a05fcbb06520c940 : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/eeef70acc63a41fd8ed60ed80f5712b5 : FAILURE in 37m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8b65f9fbf75248dfb92c5865e14bf17b : SUCCESS in 59m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de4d1b4ebd3d41b7b7dbdf9be4324830 : SUCCESS in 1h 27m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e505d795169c4e56943f27ea6b0e3114 : SUCCESS in 1h 09m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/981b59f319d74e45bf78c43afa0ced9d : SUCCESS in 17m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/83898a4ceb054df385d9788088c37e32 : TIMED_OUT in 41m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2018bc2b92be4da294cd25ee3849639d : SUCCESS in 7m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e88be959bbe4ff9823f31f5f2236715 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e063d5e5299641b086651e0b278663f2 : SUCCESS in 7m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/da5b08e4917e4dc0b48cb0d9793275fa : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c1d202dcfe9846fdb6afbf989a83a853 : FAILURE in 33m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/28e017249cde4a8491ecbbed83ce0241 : SUCCESS in 7m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/df36bc772c77403599d2986c3f16183b : SUCCESS in 2h 04m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/221473ad61844699bc1445573c29ee4e : TIMED_OUT in 3h 12m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0d15621453c147cda28c8bda31565844 : SUCCESS in 1h 46m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f7390e3e064f498391429d2043056d3a : SUCCESS in 1h 50m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/434eec5b489d48ff8370ce2c73152a6c : SUCCESS in 34m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4a20d02db6a84ac0921858c712ed903c : SUCCESS in 57m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/25d77eb0349f4073983b817bf14acd51 : SUCCESS in 1h 35m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/82c11c4c022e4a25b7ea94855fd13140 : SUCCESS in 48m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/81cbdaf0b9e94ad7a357223237292889 : SUCCESS in 1h 27m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/59799ae0402a412d872de4c652d35927 : SUCCESS in 1h 38m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6efa654ae051448e85a2437276bbcdab : SUCCESS in 1h 40m 58s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/566ef506cfa145ff8d77c6e2ba9a2971 : TIMED_OUT in 2h 03m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c760b451a414388bc99bbb5114cc781 : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":1},{"id":"791a6cbc7dde801ffbe69ebfc665eaacc93320da","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-17 22:13:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1b112f46a3c54efb9a9883bacab66d8f\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1da68f6054c94c2cbb03ff2c3b897d9a : NODE_FAILURE Node request 200-0000079008 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ddf46c49558d4cde989a5152abab26d0 : NODE_FAILURE Node request 200-0000079009 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e75d363de72647ec807f2e5b060051e8 : SUCCESS in 56m 45s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5b7fa3ab2feb411c8f7ed74d69d126ea : SUCCESS in 57m 11s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5d6cefc30bdf48d3b3a70306daa1cb30 : SUCCESS in 41m 31s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b8938ac6e03422da1f91620216476f3 : SUCCESS in 42m 03s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/80b35378fc9f4d83b227d1f4955c62c5 : SUCCESS in 49m 58s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e3d9f2ea6493485e989900086a89a7cb : SUCCESS in 48m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c1abb7e869407ee25b31f3945e9c6f8e74901221","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-09-18 10:43:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f51bbc62721ab855345aaf131baaa18726a33c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 12:08:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d7821b272948b780bc95ad984bcae4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1ba7c359f08470fb2185948309ef470 : SUCCESS in 28m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a639f8cb9f38462298926ed1ce30d570 : FAILURE in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a261244b29e7ef756e956a48acf65ec7ff5a861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 13:35:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1daaf2afbf224248a6e422556ec92299\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5185332a07824b5daf8e03c2e7dda075 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2f1f54efa804af4aa320bc991725e5d : SUCCESS in 11m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f1f6cb34132c4c559882975f0677ae19 : SUCCESS in 12m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c782afd432444add86e5ca9e6e45ec99 : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/c371439fe4c348ffb987fd7547345675 : SUCCESS in 45m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ecde2e816366489594913a5a04d58c72 : SUCCESS in 58m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd16b33dfd434b3a936a2a7b3e696292 : SUCCESS in 1h 18m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/302c25fdbd2b496aacf0d6ed3d689afa : SUCCESS in 1h 07m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5645a7402da948a591efe3212c86f5d2 : SUCCESS in 11m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f7e4cd005797454197cf64335262bb25 : SUCCESS in 21m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/40987eac27b2475689cf3415ea3082c0 : SUCCESS in 6m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79d9d4872c734282b2243d6a6a714f48 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/031e3a7bef034555a7b861cabd608a01 : SUCCESS in 8m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8c9eac68d66a4765889af8d3480f896a : SUCCESS in 4m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0b4b4f3fce12469aa3228e7c6d491f4b : FAILURE in 38m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e8b51c3f4600445daa3a0ed3444908c4 : SUCCESS in 7m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4de4941f706843ccb112f913d79e0106 : SUCCESS in 2h 05m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a7950dbbe954318827e4aae6c0165ce : SUCCESS in 2h 39m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/849b719f33eb4f418ac408f7205a0067 : SUCCESS in 1h 30m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9f1f85ffda8744af8353b229af51af56 : SUCCESS in 1h 54m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2ccc19859d8648e891bfe8d850062061 : SUCCESS in 29m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a3d343cd50b84ab0aef3d3a0f80e3bec : SUCCESS in 1h 03m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b7bcd2c7578e43ef94bcde6c10335ce8 : SUCCESS in 1h 39m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/de6d9a3f913b4d82bf8073a4ee04ea67 : SUCCESS in 45m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c5ea3f5d35a8476586921eb140be788c : SUCCESS in 1h 40m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/092315ca450944b39eafcb5aa1d4c5cb : SUCCESS in 1h 11m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/efecfd4a79574a4c9e8256b95ad7a4a9 : SUCCESS in 1h 35m 58s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ba99ced643b472ea398e5c1b36f8b87 : SUCCESS in 1h 36m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5073998e4f5441bb7f9e9384e784513 : SUCCESS in 52m 12s","accounts_in_message":[],"_revision_number":1},{"id":"66093f85888ad03c2294238b8b59d4f2ce61ab59","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-18 15:21:10.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0b1528b2a5774eb39f900e93f6d1a022\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/77f6e33a646d4e63817d44a020a77272 : NODE_FAILURE Node request 200-0000079043 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f5172f4993b4c8b939e6bd545be1455 : NODE_FAILURE Node request 200-0000079044 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fcafc8b6757d4fa799eb22978c2d81a7 : FAILURE in 1h 04m 28s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5abd7c7a47014fbb8a82d4955595f0d6 : SUCCESS in 1h 15m 26s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/33a97ac5518c413f99e0bba31a5e2a34 : SUCCESS in 1h 38m 17s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c1f16872f2b748879963742ff2b8f473 : SUCCESS in 1h 38m 18s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7124362b1ec04b849037255cc44db9af : SUCCESS in 55m 50s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/374158e58c93483c9c75c095e010a3fe : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4a6f991a28b4eead2c52bf53aaee3b770736c544","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2025-09-22 13:33:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b990b30770ee8c87190070d7b75ca5e7ba0fabab","author":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"date":"2025-09-26 21:04:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bae7b2d73691d7420f16e87d2b90f2690d4fb5b6","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-10-22 14:29:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d191d18f6111f27adc715c1594ef0ce79cf46f83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 14:30:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebd386255e813623e4280335341c81aa9c153a7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 17:37:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1268bf7272c4ca489d45186f7291e69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2afd427f34c4694801002b05b2e2e90 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0501dc76544a45839f49f6ca64092bc3 : SUCCESS in 12m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7cedfac93b0045a589f2f93bc9282c8a : FAILURE in 14m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/323867b100a04fc897d2d6b70f7cf828 : SUCCESS in 16m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/720f9224593e4d7ea948cb7a5fd79461 : SUCCESS in 43m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4dd169758a5d42c091cf6fc07701c931 : SUCCESS in 44m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bce6e5270cec4248821e67c8b89bd637 : SUCCESS in 1h 31m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9105d64a8a748ca8f1245d574a05525 : SUCCESS in 37m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc05c526cca94b0386ff9f3edd2eb8dc : SUCCESS in 17m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/36ddfd63edac4e23a58fcf80a0d98e70 : SUCCESS in 52m 51s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/09f9ab1b464d4f40b4a441eb2ad80a16 : SUCCESS in 1h 57m 01s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/476b796f739649fc8c82dc4002953175 : SUCCESS in 1h 17m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/671e2b4a704343128a8eaea4de98db62 : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":1},{"id":"129121c147916e651de27eab18cc0d4246a44bf2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-22 21:24:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"877121437f13e391ee552d389c5da013f809d09b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 23:35:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de9c6c4328eb48d48be01566dcedb254\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3dda6beb5174d02820e6bd1513e8052 : SUCCESS in 35m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2210b823515b40baa88c7c22e2a6bfdb : SUCCESS in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"baaeb77211e8dd848d7514e64ce6107c4f269a08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 00:35:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c56ad41974ce4cf7b0805f99f62602d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ea36e160563426480d946061a81fb1b : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd8b6781e27a426391ec887190865795 : FAILURE in 7m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/098ff2567f43495ba52c6ecfdfa09ba7 : FAILURE in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b066c1e781c412cb02a65da8fefc33e : SUCCESS in 7m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/14ed0e6feb934c59ad29cfd3b122dbc4 : SUCCESS in 34m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d813c260f614559b61f1f4ff415cd86 : SUCCESS in 1h 08m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a06754e0cb664ea2955d83583b89b856 : SUCCESS in 1h 14m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9da076de6ff04964b2a672283251461b : SUCCESS in 33m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1c35d90e135450f9207efa1f9658d10 : SUCCESS in 14m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6f66aed590444377b249e683eec461e6 : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5a85b9aa3879485da57808aea17c9083 : SUCCESS in 7m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a7ed621c371f474dbadc67b56b65c4d2 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/149eb1453b0e452e8d47b7bfec70f785 : SUCCESS in 6m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/908a4e2f342d439dacfc7044141298df : SUCCESS in 4m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c461b1ad24904b26bd0d368bf78a3d44 : FAILURE in 21m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c121adcf3b347c38165a73029e51ded : SUCCESS in 3m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/79f8bb1799ae4d60bbd4e8b18653ba83 : SUCCESS in 1h 56m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/290b78fd589a4840b39a067fb17f1592 : FAILURE in 2h 44m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a65668085f64b859e6cba41121d0c32 : SUCCESS in 2h 12m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4f3a3df354c64a1ca12c60a936bcb242 : SUCCESS in 2h 00m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7c6d7bb7026345148fb95725293d0620 : SUCCESS in 38m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f79a84df94a74827b482fba41451dc3d : SUCCESS in 1h 07m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3ba2a066363d4e30b0b4fe0240bf36d9 : SUCCESS in 1h 33m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c159990b53994db28c6ef5fa37cc1f3b : SUCCESS in 45m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d3d00f7f964a4289bf6bd0d35f1678df : SUCCESS in 1h 42m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f8dc8835c7c347a183f99d3cf57d110f : SUCCESS in 2h 17m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f137a001af8642c0b7dfecf4c3c72dce : SUCCESS in 51m 44s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd89375335904c8b907699399ebbdc97 : SUCCESS in 1h 35m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a1037ad89e54d4d952f0a2ed959e5da : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":1},{"id":"66867f5559be0326bbfb4ba9f6c2f5ea44de9e63","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-10-23 05:16:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdc52559693f78cfe221ac9c3bcfd220be8e8530","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 05:43:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd66cba4b39544f79d1342e5fb47875a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4672d9fee21b4bde97e28ff1a4a6c890 : SUCCESS in 24m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6737b3024bc4451ba7aabeb7de3088e : SUCCESS in 24m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9429548adc1569cb216709600e3e74cb2e657aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 07:47:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef52660c01524d458fb5c77bb7873dfc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d3937f9ef154e7391d9085c1d00108d : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cf35b9b371646f0ae94aff1be29db24 : SUCCESS in 7m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/921ed68ab0ee47bc8c7942d79ef28ba7 : FAILURE in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5aa572196444feb918e76b9c1493453 : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/9888437c643d423eb71476829865efb7 : SUCCESS in 40m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d6f6535c98744eaf82b4d8c6f814e59d : SUCCESS in 56m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a10c9d67915f41bd8a888451a4e51925 : SUCCESS in 1h 14m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82b0b1537a6449c7995dff0483101a82 : SUCCESS in 48m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76cb1899cb7a40808259f1f4b1179e44 : SUCCESS in 18m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/db9272d23c7846e884eaee9c9c29f146 : SUCCESS in 15m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9fe2a6b97c744adc834b5b7c4356ae91 : SUCCESS in 8m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1b2e31f2204d413287ffb713822b2410 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3974e9a63ffc435cbe9fee9da8ca5b47 : SUCCESS in 7m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a60317178cd94b7da778e1eb4bb5c62d : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d90a7e6b95354d0897c5d08cc8b9faf5 : FAILURE in 26m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48b9fffe0b5147a394a061c8f5b4a576 : SUCCESS in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a6f5bf72d2f14c73850900fafbfdf0e8 : SUCCESS in 1h 00m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2f8434d17a424583ae88478d1b5d1af6 : FAILURE in 1h 38m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f441e986931b4bc3b2f24135824f018e : SUCCESS in 2h 10m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b0eb85a3b7694676a5f3989fb47961fa : SUCCESS in 1h 58m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fbe17a722b50417e933a1d648a8bab4a : SUCCESS in 34m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d5ac721a8d5a4aebb2704aac6d77e785 : SUCCESS in 33m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/760b05c813e34b3a825354eb759e49db : SUCCESS in 1h 24m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/da1b6c340a2b4653aeb89b08257f7e17 : SUCCESS in 21m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/100d7968d0cd4f70b5125c89b9fbf02c : SUCCESS in 2h 06m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/aef23beb139f4cb08ce483d13391af7a : SUCCESS in 2h 26m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b363f023dc0446a5a3b1eb4fad1d2866 : SUCCESS in 1h 48m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e36074a75f934696909ff3b748f5ce96 : SUCCESS in 1h 51m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/061d447e96cf41d3932304a9154d3cf5 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d1ab78f95641a6ba2c87abdc796ccad63c6c7612","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-10-23 13:18:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b61e277eb6160d6121c5659fdb2a1f1edc4e905","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 14:00:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0facaeec59b743aab8b3e846c5d4a9dc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a05b62934dd48bfa2e667cda885b7d8 : SUCCESS in 33m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/83d6686c3d5345e49644208b80f4f933 : FAILURE in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f8c4ec73b265ca5217fa65d2d752146a63d8cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 16:23:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/992d1a75a83a442e8e9e95ce8f8b37d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f9fec4797384ab6b4fec9c76de8035b : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44680e7755e24a12af8f9b47adaa7fac : SUCCESS in 11m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cdd63df6f64407ea08daddc4f96fc3e : SUCCESS in 12m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef628f3309a54d0f828c8179e2deb514 : SUCCESS in 17m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/49fba70b184e4b23972e45ccfdda3dbc : SUCCESS in 57m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db81f5de5ed0480abb6c636c5ca9b294 : SUCCESS in 1h 04m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b10fea74ed7432ebf9f9b764c8cb4f2 : SUCCESS in 1h 29m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e5ac14f4eda4763add3e872177ef8d8 : SUCCESS in 37m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fd91aa025ae40988d99829b3801e1f9 : SUCCESS in 9m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/df2e05760f214074ad88cbb7bb239a20 : SUCCESS in 26m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eee176e0832247dfaba40fb0b5b72c18 : SUCCESS in 7m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2b510bf8165430f85b9d12fe597fe65 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c590ad5b1cee4ba9ab9eecaa8a34ecf5 : SUCCESS in 4m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f18f23ef0e2c4c37ac8bddd1e7b0505f : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3d250b8a75e54e949ef75ba5242ab636 : FAILURE in 25m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c1f2f42bea004c7cb1daa608ca60b997 : SUCCESS in 9m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0b2b729f2fd445129dbc03e0e05ec707 : SUCCESS in 1h 09m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fc139e7256024566ab744f77221fffe8 : SUCCESS in 2h 27m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/370f946a0eeb474ca45e6318d3e34ae9 : SUCCESS in 1h 49m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/917d74763e9149139a9d4d604b082b23 : SUCCESS in 1h 42m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d5016bd1a46542788cce11a9899a5803 : SUCCESS in 34m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/169fcd95ca8d4618a0a339ca087b7d2e : FAILURE in 1h 26m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/666766fbf1fa434b885570e408bffc92 : TIMED_OUT in 3h 03m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/24a1f3f5de2c40998728d83867410f48 : SUCCESS in 42m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/29362c22ea764202afb8c492ca19b0ca : SUCCESS in 1h 32m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f1c725ca20b5474bb657a6a3589deece : SUCCESS in 1h 41m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9241aa49c979480b992b3763161f7aed : SUCCESS in 1h 03m 30s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bedfa6b18e98472aa56ef30ab76b5a90 : SUCCESS in 1h 12m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f646c206fcc74a5f8486888c2ed1420f : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ea7aec2b2c1ca0d5f82a69816b6d97eb5f3c4e9b","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-10-24 05:38:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d568616ed53af134391da22d190819faa802e9cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 05:38:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be81347a477b4189b0c584a686f36814\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e01fb3f3af8b433bad86fc5d35b9eddd : NODE_FAILURE Node(set) request 2087db05951b4c7eacee9e1fe2dfd1cc failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66dd64fd7edf48e8b153ddecdc535f03 : NODE_FAILURE Node(set) request a65cfc2d0d244152815fe94268a7579c failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a94716bd020b841eb460818695a357d961e0179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 08:48:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/312ff50a34e44d9db74cb4fc53c77556\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74ec007630104d549ff1423502491884 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26f9eff79f89458ca02fe73652717668 : SUCCESS in 13m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9f693ee37e54157a7bed3c687363195 : SUCCESS in 12m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea196e9caaf4c81bd7977857bbce6ff : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/1448db0cb6654da2b0f6ec608b41d252 : SUCCESS in 53m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4e7f8973e0ba4383b0b49e24d977ec81 : SUCCESS in 42m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c647156a5fca46e999ea3dad453dfcde : SUCCESS in 1h 20m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ff349f54b344e1d8ff229e97bcc4082 : SUCCESS in 1h 03m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7778d3d37cad4c1cabd1838298449bba : SUCCESS in 10m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5481ccb62e75491da5f5e39336074e5a : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/363be7c3598c4e76b1b3caedfd538aac : SUCCESS in 6m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a172b7c3b194660b07db1d6a6936fcf : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fdfb8568251740ea90db1251a31ef3e8 : SUCCESS in 6m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5cbbcf550e6e4d8ab7782a8ac12d1b8e : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7623fb2a336e4b849fbe2d4c2e53f4f3 : FAILURE in 37m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/98ef260ace2b4f93b2db5c5a03941098 : SUCCESS in 6m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fa735a9312a749779bc57a46973f80d5 : FAILURE in 1h 15m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f9d5ce31e33f4107877a8e114cf0bdca : SUCCESS in 2h 28m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/096125a9f5e8441a952ae1ed8cc5a4c2 : SUCCESS in 1h 31m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b7ff45d8b16840828e17a664c19bc1ff : SUCCESS in 1h 52m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a34172b4885f4f44b95e62413135f697 : SUCCESS in 35m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3dee2ee052ce49a8b26a0dbe9d237e5e : SUCCESS in 1h 08m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4aea2084db0547c49c3dce96d1506413 : TIMED_OUT in 3h 04m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/04a7a22cd6964d40ad76ddf728274810 : SUCCESS in 23m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0fb86af0a4924349ae2126fb3ca31a7f : SUCCESS in 2h 01m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6805a941e20c485b953c250fd965cf48 : SUCCESS in 1h 49m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/68c9e247e473411aa22937b94c4f3843 : SUCCESS in 1h 43m 20s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4e0caf148244972b6d29e4c53024c3a : SUCCESS in 1h 20m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4fae9e92b5d4681bde1e3b62a79b6e5 : SUCCESS in 1h 04m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a269ce3133b2ffc518ac8ed16b61e7af777bb26c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-25 13:45:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"579b8fc220e53b4f7f876708e346e2e2f0abad57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-25 14:05:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c82b8893267143ada53d1ce09e78f64d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f50804e34b62475197453f7417b53d41 : FAILURE in 18m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d6c143b4695f4414bd8dafe27971614b : SUCCESS in 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62575002c462813ea3947c5eca02ec2132aaa96c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-25 15:54:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a15c57fad14d431c8e6ca06bef33cd25\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c9c58cda2274ff8b92cc2cf000a381c : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e956d8df5fb4a28b890f99d8f5068eb : SUCCESS in 14m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6672ce6c49547ef8dffca2658efbb16 : SUCCESS in 13m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a94fad1fbccf4e1da2c3439086c1df11 : SUCCESS in 15m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/2fdb77c7594f42aa9b4ea2f5426aa1bd : FAILURE in 20m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/91d5f006dc614442bbc4f949c6656804 : FAILURE in 10m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bdce872d91a641afbe9765e128dcd6a9 : SUCCESS in 1h 14m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5b8ee565c2a49b9955b01cc6ac08cfc : SUCCESS in 51m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dd48f4b3429452a87e39c66d7f744ef : SUCCESS in 11m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c19f13c8eac94787baa7158ada967f40 : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/50055fbcb5524ee18cdb99393e3a7995 : SUCCESS in 5m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/89b9c77e80a54053912c567cb26e1579 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0909176c291843d2a655bb12bc846682 : SUCCESS in 7m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/74cfaede65224d3cbd6a3cf0ea2b7b34 : SUCCESS in 6m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/10ce83dad0454c6dabbc8b05eb17c596 : FAILURE in 28m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83cea8cbb50a45c8a9433bbf27ab4ab0 : SUCCESS in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/987af10cc4654ac18dacf913431ffc27 : SUCCESS in 1h 29m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5ba0777e1db64ed7b9278470c84b418b : SUCCESS in 1h 59m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/718218500ff34961af352f016a5b4005 : SUCCESS in 1h 19m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/93c1c7ca0a004e17ab04ec02a1dc61b6 : SUCCESS in 2h 04m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bbd2d574dfde4ff3ba0072012a9422fc : SUCCESS in 33m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/776dc0fd91d5417bbc2fea6c9170026d : FAILURE in 28m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/178f70a6f7d9418d894fb1beae624f91 : SUCCESS in 1h 30m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5b5b29d652314a6280110c4ff00843d9 : SUCCESS in 40m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ea04ab8dd0f34caf921f67ea637bd5ec : SUCCESS in 1h 54m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/359a9f1be6a1440bbe9155925aab5843 : SUCCESS in 1h 58m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a1dcfa6e0a745e782e4397da73cafff : FAILURE in 1h 55m 34s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34eadffb012b4d42bbb2108d60efdd7e : SUCCESS in 1h 29m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecf7f4fd04504ca5a0092511be591486 : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":1},{"id":"57bcc4cc60b2a5aec72d5aa2c5263e2f03f6548e","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-10-27 05:50:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1c58efd005e19318ad3502173a4b64d2e2971c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 06:12:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47eca2b4499048d5ab68474871a99f06\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3dfca2a64ae447c28f4c119d4cfdc2bb : FAILURE in 13m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ff3d5f6b5bf64baa9e86ece23a08860c : SUCCESS in 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e580b9571a77693ef07fb90096aa234439096284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 08:05:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f4849456a004bfca29cbf9dfc64360c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36cc6dd597764b6eb6248e87bd578bc2 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e6a6c53905d49ecaae8b6d50b6ff02f : SUCCESS in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b707437d9d344f0b9c9d5c70c70ec1c0 : SUCCESS in 12m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/880723544ff04b4c9365fef4e75d7103 : SUCCESS in 7m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/aae8d85b4769485ca5d125081184fc9c : FAILURE in 10m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c45fe66a32474ad1838d3c2dd60cab1c : FAILURE in 7m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/80f175e55c2b42dcaa1b3e6cf8208f5d : SUCCESS in 1h 03m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb346f37f38d493cb931cd0de6b8ce9f : SUCCESS in 47m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2dd69c400254b488b4f4450ca60120e : SUCCESS in 15m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ff1a3242850548c7bb7bb0a874ecb717 : TIMED_OUT in 40m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aca3f09b7a1a4b7f9fc1b9c115b18e62 : SUCCESS in 6m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ab558478d8ba42acad39b99a87844549 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c0c6e8d754654549a9a3637aa57a5dea : SUCCESS in 7m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c7d1e921c9ed495bb6654bd2e5c0b3d5 : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/92c822715e35417bbbca611e5e966913 : FAILURE in 38m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/46eba8fd7ac3452c977ed108a5a19cdd : SUCCESS in 3m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/07a4cda4746a4ac981985eda83e89355 : SUCCESS in 1h 58m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8d84340618344f04abef72bac1ab49ae : FAILURE in 1h 55m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d374a62bf26b469aba6d2eff6321ae92 : SUCCESS in 1h 57m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f1ff87d751bc46c5b434e9db859fdb8a : SUCCESS in 1h 20m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b5e2cfe49b5a483695660099693bb879 : SUCCESS in 35m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/49982f4d830348a2be7f10bfd724240e : FAILURE in 27m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1ae72c39772a4439be1f80811ef42a15 : SUCCESS in 1h 23m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d3aef69bafe64447b9f5a660c72dfe94 : SUCCESS in 51m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/19d4cfb33fc0449382ef3454097f6cc6 : SUCCESS in 2h 00m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7717f69353ba451faf96f473f34b38c5 : SUCCESS in 2h 09m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a63876f88c242e3b5ae5396f01062de : SUCCESS in 1h 42m 17s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f2498b233094d33be70f499b8482fd5 : SUCCESS in 1h 46m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f26181aa4e842c696c6b6d7bc432068 : SUCCESS in 27m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d9fd168daa67460b2ba65020b7ab9e0f03d1a539","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-11-05 04:01:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f43301f7eea22d802b910eb49d5fb271e74f1fbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 04:24:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f078124c3a314f20a1967ffa614f4b53\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a784eba4e2e4405283e009c7f1d28785 : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8fc1c78fb7f04f47b96489d2eb4cab94 : SUCCESS in 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ef5917e98cebc11b9aab0882243fe556ed7c297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 06:02:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6a2cb44ac1c41ca99a5920b24646421\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db461208a11f40ee83c2b4d4c55d2bd1 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/054e9c220090480090adc8de253439d2 : SUCCESS in 12m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8b09379b6fa4ecfbf79a665c4d238a7 : FAILURE in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10f4db67ed4c49f6996ade3640d34cc4 : SUCCESS in 17m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/fa8215c7388249e6826e8147ed892cff : SUCCESS in 42m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1d6d0b1adb434a90ba724524c9f2402a : SUCCESS in 57m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dc41ff3700214bceb1cc201694bbbb09 : SUCCESS in 1h 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b455884aa7e44223a0765bc259cef745 : SUCCESS in 49m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbe0750fedb5452682eac1748f43ac6d : SUCCESS in 15m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9fd3adf39e3c4f3fadf037f6fe8ac032 : SUCCESS in 21m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/964b77c524154301a9c912ebb5ad3053 : SUCCESS in 6m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/40ab8b6d094344a4a59cf261378852b0 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/90c2c8e3fa69455791f3a5a73db06ef1 : SUCCESS in 6m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/720e2ac2d6754843a409014e599526fc : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f9950da4bbc4b62872c57a185b4f5c0 : FAILURE in 37m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/675886fc73c94118a183c354538a5822 : SUCCESS in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/822566ae5b55482aa3aa78a9f91efef5 : SUCCESS in 1h 55m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eb9029d3ec8245c2b6b2fc4a8d3c8456 : SUCCESS in 1h 44m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/656567ea98f24f28b74f43919597c775 : SUCCESS in 1h 48m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/23c24dec05b943dabb9d1a34adbce91e : FAILURE in 25m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cadbaa2056d34a67857b65bb7835eefb : FAILURE in 14m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee981153dfc3433286b70a39821eb746 : SUCCESS in 39m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4d29aef330624b058bec9f6acbb113c1 : SUCCESS in 1h 20m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8c9157e8c816403a8f8fc71eb49449f2 : SUCCESS in 17m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a933f4f605eb4dfaa990c7d04796535b : SUCCESS in 1h 52m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3f4ed60ff23548bea57c2662c3c585c3 : FAILURE in 18m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4a2a6577d3f48948ea747700f23b405 : SUCCESS in 1h 11m 56s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67392eb3e3aa4f988f1a70b3b3e6a6a8 : SUCCESS in 1h 07m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a879e212cb8c4ae4ae9d1c555cd31179 : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":1},{"id":"85ae7e0cc26dd4c4f5032281439e9205b8c06a55","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-05 07:01:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-961431-1-fc : FAILURE in 1h 2m 36s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-961431-1-iscsi : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":1},{"id":"eff659c7009c7bd771333cb4d6d6f536ae9d2c20","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-11-06 08:13:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b52b4e16e5fb397c9d190fdc3fc9f5cf8f29ecb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 08:33:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dba98dbc4528436fb62f87f757d8d19d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91483b75061940389593aef5cd741c4a : FAILURE in 16m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1004bb9003ce40e59c2bf5e864536fbc : SUCCESS in 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c72bbfe3cc31b0579312de166a3962f67f679035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 10:22:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ea9d14b998e46599f22702e44686af9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/709cc1e97551495a9292e4591d67ba11 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2da9c03b580042e68149ce458f338816 : SUCCESS in 12m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98f43caf6efc4849a32f468eae04f9a5 : FAILURE in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be0fb03574514e62a2a3fbf2c78649c7 : SUCCESS in 14m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/bbf60881d47843e893c5d87d27c4315f : SUCCESS in 59m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee0f385a61c444689f11eff056380720 : FAILURE in 18m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/93f5b617f5b840b3810a855e5af5f00c : SUCCESS in 1h 35m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/671e7aa8071d468da291c97e9bdc10e5 : SUCCESS in 49m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a57117416c43487e8abe1d94cd196a5d : SUCCESS in 20m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d6c960e56f1240cfa3b7e31077dc38e2 : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a2e0a67173ef46e58006707874dc64c2 : SUCCESS in 6m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3ac6c2a59df248b7a82eeef64fd430a5 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/726c2345c1c04e2c8d09a8ecc0570b7e : SUCCESS in 6m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0b9df85250e348748c64e69ca0a1b1d7 : SUCCESS in 9m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/202d666a0c7641ab89e55357a5d35652 : FAILURE in 36m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a67272e2db8b4c909e4ced27f44b4bf8 : SUCCESS in 8m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/872e5149fcaf4e9fb028e8ea2f7886d1 : SUCCESS in 1h 42m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d43c09915fdd431ab8d3182a3afb3e42 : FAILURE in 1h 30m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d9b1b410636a46e8a3d42ed75dab43dc : SUCCESS in 1h 42m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/224d64ea93174a52b81f1215515f25c5 : FAILURE in 46m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bfdbba1f2de148fda91159af8df3124d : FAILURE in 27m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/207cc7cce0d3467192a770e8fcda688f : SUCCESS in 1h 11m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c282ea113c734ca28aa0c3b5b985be62 : SUCCESS in 1h 15m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/25e2e75243ec487892e8bf10ac8e4397 : SUCCESS in 30m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7351b2fa21684788a29870e3050072da : SUCCESS in 1h 57m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6c1e9bea628a4ec58d814642e60f2058 : FAILURE in 33m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e3d2141efce14c0fbf2d3f9480491cb2 : SUCCESS in 1h 38m 23s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e6346e7a36d48dd8d48747decb7bcf9 : SUCCESS in 2h 02m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00d2d2abd1bf45749d7f664ccebdd150 : SUCCESS in 42m 33s","accounts_in_message":[],"_revision_number":1},{"id":"bd9c867bb44fdddc715b3ca31e629d1a0d8de9cc","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-06 11:12:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-961431-1-fc : FAILURE in 1h 5m 32s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-961431-1-iscsi : FAILURE in 1h 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"841fc367580c9a91b68b19e56ebbcf73a18d48f0","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-11-06 15:39:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2fa28c93cc38eb5c3d5a8a4faef38a2221d5d23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 16:05:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc6693fc8b274d0c8e6faf073eb83c52\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4a8a9d3e3b74e3d86082f578515087d : SUCCESS in 22m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9805eddb9e524071a1809b8a84f23eef : SUCCESS in 20m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2fee478b58652a3579a5279cae3eb90cba2c680b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 18:20:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf33a85b5f6e46ed9231bc1163e94c31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c534999ba6e4023a6dd38b86cb974eb : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be7113760b6c42159d4620a38288f729 : SUCCESS in 12m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a6bff54a386435d9e60ee88cfa849bf : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7032178d43478fa2ceefe21027899d : SUCCESS in 18m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/3d4e4570d1a143e38cbcad71149166f5 : SUCCESS in 58m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c348b378535b4d5ea019dfa22c87c3a3 : FAILURE in 12m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09ff1890da6f46ce9cd05859d26e791d : SUCCESS in 1h 27m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01a74e6c95134a0b814b4e4912384d2d : SUCCESS in 1h 03m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/202abab969624d3bb49f6d16fe50366d : SUCCESS in 14m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/52e0b8bd6f2c4b5890f615fac3d23f69 : FAILURE in 23m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6a037c274cf0441198e116c6d41f2420 : SUCCESS in 8m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc0765ef4c484e52bd59041fc53436ad : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1cf254292e147638fc1f6bc6fbc737c : SUCCESS in 7m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a789781916454b12a0704fa35813332f : SUCCESS in 7m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/97c1554c4fa04455a4032a6a6276c9a8 : FAILURE in 38m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f4b445b6b88d4b928058cc5b45e64742 : SUCCESS in 8m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/29a105aa9c604f55b8e81413c99d1dc6 : SUCCESS in 1h 56m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48b930f4da414874a041251ef287702b : FAILURE in 2h 34m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cbf753d69cdb4cf2b4781356500ef838 : SUCCESS in 1h 55m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/11b2313aecb74e81b956d374a97ed996 : FAILURE in 20m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/371afdfcb49d43fd8f8c5fd03c994a05 : FAILURE in 24m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ffd462518ed24c28929ce4f5d6a8bc6f : SUCCESS in 59m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/126b8143d95b4aa4859ca8cdcccdad9b : SUCCESS in 1h 21m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/05398ef56fa941058d994ae3da4785ef : SUCCESS in 23m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/898c51eba71e43cbb38f2e222487cf11 : SUCCESS in 1h 49m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7fd54720949344a290dc42e502af47e4 : FAILURE in 35m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e7d6de1b8f94874b9c2774cf75a50a3 : SUCCESS in 56m 28s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6379e31e15524d45a64f2eef49899dae : SUCCESS in 1h 43m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee8883c0bcf44a49a6adbe3bf2d75ecb : SUCCESS in 33m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b77d40c9b119d62def29ce273ec6a0518ffa8fff","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2025-11-10 14:33:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"503ba46f7adb35f9695137a1d6302fcd15dba65a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 14:59:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0781de8dedd548e68eaaea5527bd82ba\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f71ef0851895445086bbf94266534fa4 : SUCCESS in 19m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99c21badb35f4de4a2ccbb03ca339f1d : FAILURE in 19m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"433d879e236d7a3bbeb5ff6eca3395114fcb8435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 17:17:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f921a7ac96244b1baa4afc9d78698352\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e88a5970edd4f7c983a82f4bb67979b : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e53e392ce074cb79a7aaa9bcde2d3f3 : SUCCESS in 12m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85906e69a0ea468da698b40c146c2276 : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5b0dcd65a74c439caeceee1ef0cb49 : SUCCESS in 13m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/a33032f565f342cc892356c10bfa3753 : SUCCESS in 58m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/02c14e8efdd947da8ecd56a3566aa1f3 : SUCCESS in 1h 00m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6704a8c7c9c24456a712b5b12ef8e757 : SUCCESS in 1h 08m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/158c3394dea94a54942c2c08dbb4c188 : SUCCESS in 1h 03m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c320119d85f4290bfdb487b5667e306 : SUCCESS in 12m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9a9bf65853e241ba9b20437487c6d7a0 : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6b5b1e95eca44fc7bc154088d26547ea : SUCCESS in 6m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e8b01428c72e4044a2f90f250efb966d : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f62ac819b01b41c4a580d2ba37c97d9a : SUCCESS in 6m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7399b04a7c4a4ee58cce929f9e5a0a94 : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1c834b2b9adb445197c24ddc47856b39 : FAILURE in 26m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/49d8f36a556a4720b6d22b38518f3832 : SUCCESS in 6m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e9e2302513f044c8878919c0463822ef : SUCCESS in 1h 54m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0f974dfdc2a2472fad180c05b9e2754f : SUCCESS in 2h 37m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3f667c19f4dc4e88b8d3d7338ac09f22 : SUCCESS in 1h 31m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b822a388b66844c3ab35226979796d7f : FAILURE in 34m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/475d577dfe824a5bae3aff974df79c92 : SUCCESS in 38m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5971e3930da24bd49cfa9a6f9af29646 : SUCCESS in 1h 07m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/07e83a1b2ec34aa5b466bcaaaa68db68 : SUCCESS in 1h 41m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a948e8d0ec954091b80c41200cad17f9 : SUCCESS in 43m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7ee52ec446d545d5a8a05dc0000a7c56 : SUCCESS in 2h 08m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/45c248849de940f39a15e17ce7dbad60 : FAILURE in 24m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e9b41004ca8442d9f8db257eca12fec : SUCCESS in 1h 48m 06s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d352675aa4b4027a9d738f30603456c : SUCCESS in 1h 27m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa83442b50594e19bfe630a5293699ce : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c2fbc7e033dc453d35db4c928b84e237738b76e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 17:17:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7909e7efc237388e297d83e46f2dc83663a2009","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-10 17:41:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-961431-1-fc : SUCCESS in 1h 5m 36s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-961431-1-iscsi : SUCCESS in 1h 11m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ef4a330ee32589a196ef9d27e19c59d96beba020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 19:24:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a10a289ec4b64914b03bb8ba3d71d298\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bb53390feff4899a984d34f6190c6c6 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e14e6cb7748a4cc8b1d9a0fe3d7b609f : SUCCESS in 14m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9331d77080f9464eaee86b3de15d591f : SUCCESS in 12m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ab76f56ba24db789ec0a1571874923 : SUCCESS in 15m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/44b392cbeaa244458c8079514115bde6 : SUCCESS in 53m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1f8ebae977004e0a8e1a4dfed1c968df : SUCCESS in 31m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/041f8fa883314c7b8e0c1e8602428be2 : SUCCESS in 1h 47m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a652202b9e1456cb0b6d00a6c8836ab : SUCCESS in 1h 08m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8eb19129a4ea41e7b8b18bbb68be8034 : SUCCESS in 15m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc0181f31fd842f68b49736017cd3db0 : SUCCESS in 1h 07m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/81c3b99a596d4a3cbbabb0d176ddf9dd : SUCCESS in 2h 02m 18s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00bed2c5672f4562915fc4e01a817025 : SUCCESS in 1h 43m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/071a198875f84fc18a374ea98ee65c3e : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"28ca8a6ff6a35289fb7dab809c70b89ee78da3ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 19:24:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1f86d457a072221ce0ddf4438e4b2bb5e4decc15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 19:25:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7c315144874e85bf5a72c1fbefe9e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2cb3716303a4c27b40526f02a5bc6e8 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b49ee66397e46c2b9559d59cef6ce7b : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1a0af2969be2d8e3c640033d7a2c2db6fd42142","revisions":{"b1a0af2969be2d8e3c640033d7a2c2db6fd42142":{"kind":"REWORK","_number":1,"created":"2025-09-17 04:17:29.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/31/961431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/961431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/961431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/961431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/961431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/961431/1"}}},"commit":{"parents":[{"commit":"677cc07bb5ec96f917e6afeba5da13214be24d84","subject":"Merge \"RBD: Fix issue with managing volume with type properties\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/677cc07bb5ec96f917e6afeba5da13214be24d84"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-09-03 09:09:52.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2025-09-17 04:17:29.000000000","tz":0},"subject":"[NetApp] Enabling total_volumes capability support","message":"[NetApp] Enabling total_volumes capability support\n\nCurrently, NetApp driver doesn\u0027t have a way to filter out the backends\nat scheduler level once maximum number of volumes is reached per pool.\n\nThe total_volumes capability support is added and default function is\nupdated for iscsi/nvme drivers to filter out the backends once the pool\nreaches maximum number of volumes w.r.t driver limits which is 1024.\n\nCloses-Bug: #2117263\n\nChange-Id: I40263682af4735406d341e77ca90ee37cb361994\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n(cherry picked from commit 9c63da02a4573e51aefc9f629d3553d60518a394)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1a0af2969be2d8e3c640033d7a2c2db6fd42142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1a0af2969be2d8e3c640033d7a2c2db6fd42142"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
