)]}'
{"id":"openstack%2Fcinder~964137","triplet_id":"openstack%2Fcinder~master~Ica02778db1a2ba90e41c25de06a9850b4267bf06","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-04-15 14:31:37.000000000","reason":"Change was abandoned"},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-04-15 14:31:37.000000000","reason":"Change was abandoned"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-04-15 14:31:37.000000000","reason":"Change was abandoned"},"38411":{"account":{"_account_id":38411,"name":"Hemanth C","display_name":"Hemanth","email":"hemanthc@netapp.com","username":"Hemanthc","status":"Netapp"},"last_update":"2026-04-15 14:31:37.000000000","reason":"Change was abandoned"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-02-21 04:45:59.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}}},"hashtags":[],"change_id":"Ica02778db1a2ba90e41c25de06a9850b4267bf06","subject":"[NetApp] Changed the path parameter to relative while triggering file copy REST API","status":"ABANDONED","created":"2025-10-15 04:59:41.000000000","updated":"2026-04-15 14:31:37.000000000","total_comment_count":7,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"aabd26f85aea07cc22b66e894f48e2dd93e80402","_number":964137,"virtual_id_number":964137,"owner":{"_account_id":38411,"name":"Hemanth C","display_name":"Hemanth","email":"hemanthc@netapp.com","username":"Hemanthc","status":"Netapp"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-27 13:27:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"disliked":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":-1,"date":"2026-03-18 15:31:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":1,"date":"2026-02-21 04:45:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":1,"date":"2026-02-20 06:49:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-15 05:00:34.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-15 05:01:26.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-15 05:07:50.000000000","updated_by":{"_account_id":38411,"name":"Hemanth C","display_name":"Hemanth","email":"hemanthc@netapp.com","username":"Hemanthc","status":"Netapp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-10-15 05:07:50.000000000","updated_by":{"_account_id":38411,"name":"Hemanth C","display_name":"Hemanth","email":"hemanthc@netapp.com","username":"Hemanthc","status":"Netapp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2025-10-15 05:07:50.000000000","updated_by":{"_account_id":38411,"name":"Hemanth C","display_name":"Hemanth","email":"hemanthc@netapp.com","username":"Hemanthc","status":"Netapp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-10-15 05:53:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-15 05:57:17.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-10-15 05:58:07.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-15 06:00:10.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-10-15 06:44:29.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-10-15 08:28:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-15 08:35:23.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-11-26 06:57:28.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"CC"},{"updated":"2025-11-27 10:38:44.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-11-27 17:30:45.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-11-27 23:48:23.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-16 09:24:21.000000000","updated_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"reviewer":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"state":"CC"},{"updated":"2026-02-20 06:49:47.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-03-18 15:31:28.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"5d7985d44a688e0235c3d4ebb405ee538ddf9300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38411,"name":"Hemanth C","display_name":"Hemanth","email":"hemanthc@netapp.com","username":"Hemanthc","status":"Netapp"},"date":"2025-10-15 04:59:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5211e83c16335f590be016d13cb11acdcf83933c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-15 05:00:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"3a6419e27ddbf5f6db254d7523ffe533c897d80b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-15 05:01:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/34d024ad5ea54266ae5b64cbca53c3d6 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9862ed2475284621ba1c6a89ab9c18f2 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"3c6d9b5ae40f3d18c72b0350af4a5800ea760207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38411,"name":"Hemanth C","display_name":"Hemanth","email":"hemanthc@netapp.com","username":"Hemanthc","status":"Netapp"},"date":"2025-10-15 05:22:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"20d8b6cdb6c4f5ce919d121a54809c1d4430f1bb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-15 05:23:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"848fcd3dad245c06761c4d65b6adb969f1dd6707","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-15 05:24:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/29019365f8a540f0bb2b3d2f27b7def0 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/281653e740494b0c8d87125f31eba313 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"8d320145143ec0f071847045f230442d69cbc442","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 05:53:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5b3fba730f74d6b877169d241b2033a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e65b826ae257476ba1c98ed9af3d93e3 : SUCCESS in 20m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a86ccfad44d444bda8cd3239f980f70d : SUCCESS in 22m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/43d88b35ba77485fa107e4c9d6ccd6ee : SUCCESS in 28m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"210550f88b6e59d04964e14d5fc32a2cd9ceaa5a","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-10-15 05:57:17.000000000","message":"Patch Set 1:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-37-964137-1-fc: FAILURE\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-37-964137-1-iscsi: SUCCESS\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":1},{"id":"740c3612510bcc784e38e80fa50e4434389a38ec","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-15 05:58:07.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ica02778db1a2ba90e41c25de06a9850b4267bf06 .","accounts_in_message":[],"_revision_number":1},{"id":"efe43a2f4b8dc698d4feb4705d747bd96489b71a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-15 06:00:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2035/.","accounts_in_message":[],"_revision_number":1},{"id":"1194cc72ca64412d858a80cc2ca6a11b2560293c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-15 06:03:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2149/.","accounts_in_message":[],"_revision_number":1},{"id":"91ec42526255e013b3f774fd7322f24a371e1af5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-15 06:17:19.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ica02778db1a2ba90e41c25de06a9850b4267bf06 .","accounts_in_message":[],"_revision_number":2},{"id":"e41c4e6aabfc83f3f3be5167c41015e108398bf7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-15 06:44:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/964137 : FAILURE in 1h 17m 40s","accounts_in_message":[],"_revision_number":2},{"id":"7f1ee861c9b94d97129f860efcd4869dd49182c1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-10-15 06:51:06.000000000","message":"Patch Set 2:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-37-964137-2-fc: FAILURE\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-37-964137-2-iscsi: SUCCESS\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":2},{"id":"4ed56014262e97cfff73a1372949ae155ce48d4b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-15 07:04:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2150/.","accounts_in_message":[],"_revision_number":2},{"id":"d85adae052a8137b8b6dd25e9dd73737445fc4e3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-15 07:59:46.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2036/.","accounts_in_message":[],"_revision_number":2},{"id":"e040a402c1bcedd5261eddfb590c110423817968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 08:28:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7754a12be4de4ab1a6b33df153ceea80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62da74a911a2424c97ba1986fdf02e29 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f1098c5efe748a0a9441db6108302c2 : SUCCESS in 8m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/643437f82254465aa99cabe756dbef95 : SUCCESS in 15m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82fa9f390e1458a9666abbf98f5e1b0 : SUCCESS in 11m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/10708108f6264acb87adc5cee7879170 : SUCCESS in 55m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61db95b51bd1443ba001f924a8a1e97a : SUCCESS in 51m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/812259579939474fa4e05656d04f224f : SUCCESS in 1h 16m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6733fba1b764ddfb781e55fe307c781 : SUCCESS in 44m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bcfb7601082422da4c869a3f66866f4 : SUCCESS in 15m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f64855f048224b4a8a52debdb2addd4e : SUCCESS in 18m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f8ca039812004f7bb2dc10fb4fdeb1ab : SUCCESS in 2m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/81f0144250cb428fbca7b6ffb55c578e : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5028eb535acc4dc2ac469c6da6ddebc7 : SUCCESS in 3m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cc3c75742d104eb5b7e7a43dd0dc3725 : SUCCESS in 9m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9d58274964094fbb8c6e8a3cc3695d5b : FAILURE in 26m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d2814c24f3464a21bfc29b33183677ca : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bbfa08b421c74fbd87bec1e6cb9ebc65 : SUCCESS in 1h 49m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e9dcc3b76dde405fb8cb47672300ab16 : TIMED_OUT in 3h 05m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6dc88a13390346dca61be0c4cd45775c : SUCCESS in 1h 51m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/aec84b54c6f6477cab97442da8bb52e2 : SUCCESS in 2h 09m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/75aff61a7dce48d8b89865c7283b50b8 : SUCCESS in 22m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d297e270e524e1a88a277e5d78b647e : SUCCESS in 1h 11m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d636930aee13402d93d4691c194c6034 : SUCCESS in 1h 28m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/73c80b3a291d4834a444529e488c50a8 : SUCCESS in 28m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c9ef2bfb70334e1eaa41e3cdd9e0e7af : SUCCESS in 1h 05m 19s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/91a5a52821a84c639c1403e25ac8556c : SUCCESS in 1h 49m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f97500811a574b33b529aa61377666c4 : SUCCESS in 1h 26m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aad13b3bcb4546b2987ebfa563a08aff : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":2},{"id":"34771d6f31115c5c16bdeb1cc0609a22b925da25","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-15 08:35:23.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_10_2025_08_29_03\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"feec9205c0971cacdaa721d87b8844d24fb99c83","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-11-26 06:57:28.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e99fcaa2ee46a6e3b4db282d0b58568342ebff76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-11-27 10:38:36.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"dd7256e18e2833933989e4d70d459bc7795b15fb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-27 10:38:44.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/76/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"3b794204731482efbe6d30af9f201de5d0e77520","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-27 11:32:39.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ica02778db1a2ba90e41c25de06a9850b4267bf06 .","accounts_in_message":[],"_revision_number":3},{"id":"c8a27da417ab61cc5ed2c8a33646944a2a4455f6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-27 11:33:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2582/.","accounts_in_message":[],"_revision_number":3},{"id":"d6cffcf99000ec57d427613308448b650fd1f9fe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-27 11:34:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2730/.","accounts_in_message":[],"_revision_number":3},{"id":"c6ca60da76c5c14c7ce35d233822aa53bcb52935","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-27 11:46:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/964137 : FAILURE in 1h 06m 29s","accounts_in_message":[],"_revision_number":3},{"id":"76d67e635019a802cda69ca9728babb5e3e2459a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 11:56:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/216e8506588741d496281a65f87dcf2d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4578890bfa604205a24fce96646a4d84 : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/05b30c5a1ff14feca6b05b8a8a22e5be : SUCCESS in 29m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/230ebf5dacfe4e70a37da985be0aa484 : SUCCESS in 16m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5bbf9d82acf10c4adf6affd829f2c395f0ef12b9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-27 11:57:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b795760efc3a46eeaba08fc4d567f2a9 : SUCCESS in 1h 15m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e9e477a4cfcb4a3e9fb489224b6d65ff : FAILURE in 53m 35s","accounts_in_message":[],"_revision_number":3},{"id":"d607525307d433428ab396f662bf8ece2f615d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 13:27:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a3e806c09ab4e4cbf03031d3aff4ff2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f202aabd8ca84629b884a5b1ab8d81a6 : SUCCESS in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/237344bb717c43f3b5542491911ee478 : SUCCESS in 8m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abbeacfbddb349b08af7b1061bcd53c7 : SUCCESS in 14m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a65e2817da24a15865d90de8c05bf30 : SUCCESS in 15m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/ceae99ebec6e461d8c4bd97e7d6565bb : SUCCESS in 35m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eecbbf14381e452b960fb20daa3c8c0b : SUCCESS in 51m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01971054d1e043fcac919e6e85737465 : SUCCESS in 1h 25m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2afa30f636943c7a2aa9bc4d9ccf601 : SUCCESS in 37m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db526c3b856843fd8d5d7a043236b888 : SUCCESS in 15m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/10eed4c1caa64236a8cc7b15ccefe57d : FAILURE in 25m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8e7f996d16a646f3bbfc8d010d4fd60c : SUCCESS in 7m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9ba1181a030140a8986735739adba2d4 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce950740e7014526858802beb9cd0981 : SUCCESS in 6m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a27b3b6ac37848ffa1d9822128b75e5b : SUCCESS in 10m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/df1f1ee6ea56406ea5ae7b9d66649a71 : FAILURE in 21m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e8a3e92c5b1f41748d7df8d35b3e1dd6 : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a2c2b0e6fb9a47f4b3657675d571c9d4 : SUCCESS in 1h 02m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3d86c4a72d354fa9917e95653ee4fe8c : FAILURE in 2h 44m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5b22ad3a604c4bfd844116b703393dc7 : SUCCESS in 1h 55m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/84148e09ce2347b29c362533c01c27ce : SUCCESS in 2h 18m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9072c77a75fc49f7a746c5aeb860f1df : SUCCESS in 18m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1911ea98ac4742e69a4456c57842a10f : SUCCESS in 1h 07m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fc204fc16e154b27af6a22756df0875c : SUCCESS in 1h 38m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c9148553fea346d9b343fd5fc7aa9669 : SUCCESS in 36m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9cc881797bed401fb9c1b162ded9eb07 : SUCCESS in 1h 41m 43s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/eb55114e9e00484bb6c5827c962f5731 : FAILURE in 1h 25m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e5b034d1654476fa305d65ff050dda9 : SUCCESS in 1h 49m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14235f2cd055493b8a53f40ec857a5f3 : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":3},{"id":"687a1a2ea56403b841480d2fc27fb51dc6c20cad","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-11-27 17:30:45.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/37/964137/3/20251127214958/NEC-V-ISCSI : FAILURE in 3h 36m 12s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/37/964137/3/20251127214958/NEC-V-FC : FAILURE in 4h 40m 44s","accounts_in_message":[],"_revision_number":3},{"id":"882fe2d3185e9710b7a513c2fe4d8f38b8c3b36c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-11-27 23:48:23.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/37/964137/3/20251127222656/NEC-ISCSI : FAILURE in 9h 42m 45s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/37/964137/3/20251127222656/NEC-FC : FAILURE in 10h 21m 24s","accounts_in_message":[],"_revision_number":3},{"id":"2cf75317112b937adf36895121b03ef7566ebf1c","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-12-16 09:24:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fa8f8de8b65177408d8202cc8da893d59b6c22aa","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-12-29 15:44:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0536b037b7a9f1d6d34bedb9dbf287d0b468e19a","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-20 06:49:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"435c94f7b2a57508b3d0c1495ecfc2463180ead5","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-21 04:45:59.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0fd481c31cff44b23d7cf064130ee8f4cdcfbd21","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-18 15:31:28.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aabd26f85aea07cc22b66e894f48e2dd93e80402","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38411,"name":"Hemanth C","display_name":"Hemanth","email":"hemanthc@netapp.com","username":"Hemanthc","status":"Netapp"},"date":"2026-04-15 14:31:37.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/cinder/+/967734 is raised with similar fixes + additional changes. Hence marking this review closed","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1e2fb708d998bda25007fccb1f3e1e39dffef049","revisions":{"64a6eab2e822003d525dc829ca472bbb6a3b9c45":{"kind":"REWORK","_number":1,"created":"2025-10-15 04:59:41.000000000","uploader":{"_account_id":38411,"name":"Hemanth C","display_name":"Hemanth","email":"hemanthc@netapp.com","username":"Hemanthc","status":"Netapp"},"ref":"refs/changes/37/964137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/964137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/964137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/964137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/964137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/964137/1"}}},"commit":{"parents":[{"commit":"eb3666620dc455a39f005b95d7f286ba17896a9f","subject":"Merge \"NetApp - Add additional debug logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb3666620dc455a39f005b95d7f286ba17896a9f"}]}],"author":{"name":"hemanthc","email":"hemanthc@netapp.com","date":"2025-10-15 04:50:38.000000000","tz":-240},"committer":{"name":"hemanthc","email":"hemanthc@netapp.com","date":"2025-10-15 04:58:45.000000000","tz":-240},"subject":"[NetApp] Changed the path parameter to relative while triggering file copy REST API","message":"[NetApp] Changed the path parameter to relative while triggering file copy REST API\n\nChange-Id: Ica02778db1a2ba90e41c25de06a9850b4267bf06\nSigned-off-by: hemanthc \u003chemanthc@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64a6eab2e822003d525dc829ca472bbb6a3b9c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64a6eab2e822003d525dc829ca472bbb6a3b9c45"}]},"branch":"refs/heads/master"},"3fb05f8b360b5deabc387aea9562f5d809980bc1":{"kind":"REWORK","_number":2,"created":"2025-10-15 05:22:35.000000000","uploader":{"_account_id":38411,"name":"Hemanth C","display_name":"Hemanth","email":"hemanthc@netapp.com","username":"Hemanthc","status":"Netapp"},"ref":"refs/changes/37/964137/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/964137/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/964137/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/964137/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/964137/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/964137/2"}}},"commit":{"parents":[{"commit":"eb3666620dc455a39f005b95d7f286ba17896a9f","subject":"Merge \"NetApp - Add additional debug logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb3666620dc455a39f005b95d7f286ba17896a9f"}]}],"author":{"name":"hemanthc","email":"hemanthc@netapp.com","date":"2025-10-15 04:50:38.000000000","tz":-240},"committer":{"name":"hemanthc","email":"hemanthc@netapp.com","date":"2025-10-15 05:22:19.000000000","tz":-240},"subject":"[NetApp] Changed the path parameter to relative while triggering file copy REST API","message":"[NetApp] Changed the path parameter to relative while triggering file copy REST API\n\nChange-Id: Ica02778db1a2ba90e41c25de06a9850b4267bf06\nSigned-off-by: hemanthc \u003chemanthc@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3fb05f8b360b5deabc387aea9562f5d809980bc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3fb05f8b360b5deabc387aea9562f5d809980bc1"}]},"branch":"refs/heads/master"},"1e2fb708d998bda25007fccb1f3e1e39dffef049":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-11-27 10:38:36.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/37/964137/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/964137/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/964137/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/964137/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/964137/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/964137/3"}}},"commit":{"parents":[{"commit":"eb3666620dc455a39f005b95d7f286ba17896a9f","subject":"Merge \"NetApp - Add additional debug logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb3666620dc455a39f005b95d7f286ba17896a9f"}]}],"author":{"name":"hemanthc","email":"hemanthc@netapp.com","date":"2025-10-15 04:50:38.000000000","tz":-240},"committer":{"name":"jayaanand borra","email":"jayaanand.borra@netapp.com","date":"2025-11-27 10:38:36.000000000","tz":0},"subject":"[NetApp] Changed the path parameter to relative while triggering file copy REST API","message":"[NetApp] Changed the path parameter to relative while triggering file copy REST API\n\nFixes LP Bug: #2131955\nChange-Id: Ica02778db1a2ba90e41c25de06a9850b4267bf06\nSigned-off-by: hemanthc \u003chemanthc@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e2fb708d998bda25007fccb1f3e1e39dffef049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e2fb708d998bda25007fccb1f3e1e39dffef049"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
