)]}'
{"id":"openstack%2Fcinder~964289","triplet_id":"openstack%2Fcinder~master~Ic6afb506ba4727746925368dc714d829967be3a7","project":"openstack/cinder","branch":"master","topic":"linstor-upstream","attention_set":{"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-09-04 07:44:59.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"33067":{"account":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"last_update":"2026-09-04 08:29:10.000000000","reason":"Someone else replied on the change"},"35467":{"account":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"last_update":"2026-09-04 07:44:59.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-04 07:44:59.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic6afb506ba4727746925368dc714d829967be3a7","subject":"linstor: rewrite driver v2.0.0","status":"NEW","created":"2025-10-17 08:50:41.000000000","updated":"2026-09-04 10:24:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f344d435d3746fffd8c6d4d4af9f69323c94489e","_number":964289,"virtual_id_number":964289,"owner":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-04 10:13:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"}],"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-17 08:51:35.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-17 08:52:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-17 09:48:15.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-17 10:15:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-17 10:51:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-17 20:13:50.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-05 13:41:33.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-21 13:42:27.000000000","updated_by":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"reviewer":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"state":"REVIEWER"},{"updated":"2026-02-19 12:42:04.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-02-19 12:46:51.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-02-19 12:57:10.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-19 13:08:46.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-19 20:09:21.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-02-23 12:49:40.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-02-23 14:15:16.000000000","updated_by":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"reviewer":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"state":"CC"},{"updated":"2026-06-05 14:47:34.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-09-02 15:23:42.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-09-04 08:29:10.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-04 10:24:44.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"41fcc1200b873111d39ab330dd81048567c7eec8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"date":"2025-10-17 08:50:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3331fda491351e8db01c442671bfd3ae898e4da1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-17 08:51:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"7a7d0721b9fda5b7487438f91604b23788308632","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-17 08:52:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b88eef18d67e40c5a8f1ae70a9277793 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dbf2705efaf14741a0621c855351ee95 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"f7fcfa0152a2f1258e926845b2f84b1d38cf0785","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"date":"2025-10-17 09:16:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"8e924e67306cfcc9fef091aed3ae139f1a9c78a1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-17 09:48:15.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ic6afb506ba4727746925368dc714d829967be3a7 .","accounts_in_message":[],"_revision_number":1},{"id":"ad18e5a200b33c057c0eeae69b3a9b805a8b56de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:15:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/882273d531da491a933e7622bb963997\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dfef2769300c4cfa8aad0cfd258f6f71 : SUCCESS in 22m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3de88007477d4b068d26d89c43ab51a4 : SUCCESS in 21m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c883c8569f1f44b7848c2d74f3e3ca44 : SUCCESS in 31m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"294be3c63ae35dba61a492e65f0dd4edb671439b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:51:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a03c1a36066946ba99c2cb1519162f5e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79a9273ee8664373a00db32fcff1dec1 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7478907468e248c68ca3fb92c1d89675 : SUCCESS in 8m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5223472633e4dd6be3a82d4a70e4ef1 : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45c05f91c1f941bba0599833fa4178c6 : SUCCESS in 15m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/cfaaf43611ff4962a13049b0c6485d01 : SUCCESS in 1h 03m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/461e68d6f2724b5f928086db4fb41ceb : SUCCESS in 1h 07m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d352abbf85eb4db8b7a55fbc37ced172 : SUCCESS in 1h 26m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/070bf0e2194f4f6d891cdb8c7d0f64c8 : SUCCESS in 57m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/aa106a4313f0499db83989bdfcf03ce8 : SUCCESS in 23m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/76df0ec82dcf4338bd535e626e076828 : SUCCESS in 6m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/121af82f0ee44d6fbcb5070f1047d90e : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9564856f0df040738b95fe27fa35294f : SUCCESS in 5m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/87f717328a3c47718480fed15ed6f7db : SUCCESS in 8m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5628cd951a1c4afd9d041bb7109e5222 : FAILURE in 36m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/35c926d01d134e80b4bfb12a16f5ad36 : SUCCESS in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d394ad7896b94666b0099c3975c936b0 : SUCCESS in 1h 20m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/04991e765cef4d00bf7b9dfc560fc8be : FAILURE in 1h 53m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e75e9c47b93b424d8c85081a578c55ba : SUCCESS in 1h 53m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b9457c61c45e43809f50734689ce5f14 : SUCCESS in 1h 40m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5f7804f98a4a483b8a1ac085c2a756f2 : SUCCESS in 19m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cdd250db07cb4050839f8a47b8a573dc : SUCCESS in 1h 11m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2d46d1afb49940458085b98e7cb99c09 : SUCCESS in 1h 15m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/23f66d9606cf4291953ee5c2d450d38e : SUCCESS in 41m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/80c6002521f04c29b2e94655fd22c74f : FAILURE in 57m 19s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/984a2e5fade8418fb2ce256d208289d4 : SUCCESS in 1h 54m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a3e1e94888694a7fad1289fa364ad8e5 : SUCCESS in 1h 21m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1664c5b68b34252aeda4be0281f9743 : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":1},{"id":"bdd8bcec1b51939775881aecae5b87486cd101b5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-17 20:13:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2187/.","accounts_in_message":[],"_revision_number":1},{"id":"bb265436f6b36fe60dae471befcda1cd83dc0cb8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-18 20:41:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2073/.","accounts_in_message":[],"_revision_number":1},{"id":"954af0166376dc39d0df40e1625d0e79ab9f200d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"date":"2025-11-05 12:27:57.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"71e04632266da8f0396f03119ba524637ee45914","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-05 12:28:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"bbd40b9d08c5d9ba3ff4b3a2edbedeb597269c19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 12:57:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93ea80727b7a4092ae0d3352d7960dee\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ca78af43df24b029c0773a78e57a3c7 : SUCCESS in 20m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32c010887c8148ee9cb0e353e1951a48 : SUCCESS in 20m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd9cb861695f4372bda6e1fc92ea4069 : SUCCESS in 27m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a0642096ca27f6ac27adf8b43279f0deb0030ad","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-05 13:25:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/463e343de1364cc591e80e4fc910009a : SUCCESS in 45m 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3b6a7ea1f54d413b82c3b125c4bb2baf : FAILURE in 51m 07s","accounts_in_message":[],"_revision_number":2},{"id":"fe5c9ff143760bfdf53c25dc6cb2607dcae0a1e8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-05 13:31:25.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic6afb506ba4727746925368dc714d829967be3a7 .","accounts_in_message":[],"_revision_number":2},{"id":"24e21bbe2e04842c8ed62b93b1b8f4366430204d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-05 13:41:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/964289 : FAILURE in 1h 12m 10s","accounts_in_message":[],"_revision_number":2},{"id":"1308b41c4e7f93c8cad81436101cc7da5a62dcc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 15:15:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06a706049b8a4a8290241589682416b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceaa686de3d6432685b25bb313a70fc0 : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3264eef0b0a4981b63d4daf2725d42b : SUCCESS in 10m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a4e2328e8024e318a8a18e37f1d4f61 : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f33e362d0684542a9fdde99c1d13ba9 : SUCCESS in 15m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/8a1116e296524b5e87796bd0c71de4df : SUCCESS in 56m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/46753f520ed7488ba2aeb4b790a347c3 : SUCCESS in 43m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/efb73a422159463c934c1029df5f6e30 : SUCCESS in 1h 30m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4978396b50a647d799a1c865073fe7d1 : SUCCESS in 59m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/626127dffa9e4e81a2513f9dffaec31b : SUCCESS in 23m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bd4af2293fdd4c4fbef628d4bc9c5398 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a5e2edf1e70b4ec18c06e26b116ebc5f : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6643477400de4d5aa9abf247e0c2a43d : SUCCESS in 5m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d8ef1ad1afda407da2030f31faaa6785 : SUCCESS in 8m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/afff9e2c5b584f9096936d35949cc44c : FAILURE in 29m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/33b6a74fc24141fb993ec2c50b1fc388 : SUCCESS in 5m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/64af28f7c29b41e785929358757c69c2 : SUCCESS in 2h 03m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9ccc78c6d0b24193b5777a8869436ea4 : FAILURE in 2h 39m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d31b48615ff2422baf6a59dc2131f5d7 : SUCCESS in 1h 49m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c8630501e5b140a99114d6b2d55b7a85 : SUCCESS in 2h 04m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c275964c6e284aeda3d6117922ebeba0 : SUCCESS in 38m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/820eb703cd1b4b3fbef4b8f789ad8490 : SUCCESS in 1h 09m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c7991e51184b4f4a917798acba8ed444 : SUCCESS in 1h 32m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f781c70f5a6344bca258c8d18bd4b6b7 : SUCCESS in 46m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4d3143cda5c24a1892cbeb3beaf928be : SUCCESS in 2h 03m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/43b6426002ab4d6f9efe58081e3e938c : SUCCESS in 2h 15m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a4db0d4019c4608a9d0b64c69c03e36 : SUCCESS in 1h 42m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24179e2bc8fa4135af7840a681daf9e4 : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":2},{"id":"5ac0cd04f838c000d843973adeeac60940287f64","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-05 21:00:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2422/.","accounts_in_message":[],"_revision_number":2},{"id":"8da9a8eae3a09035d1b105f4cfde1e5399dfb3b2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-05 21:17:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2308/.","accounts_in_message":[],"_revision_number":2},{"id":"09f2fd6590a9b0e7debe724f0557f5200d2906db","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2025-11-21 13:42:27.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48870790f7704d2d2aa3ae0533fa81c5ba48617e","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2025-11-27 18:28:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d033aaa794d6b144d6ff47840f9ff938a0e7ddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"date":"2026-02-19 12:06:39.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ccdd41525678e4c17893b22d33f7935b8f05580a","author":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"date":"2026-02-19 12:11:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ff49d637ba1c0901b47bc0c6f3c9c740ece3c990","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-19 12:11:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6889086e02d44bdcb0e759d35c201d03 : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cd7adcc030f442168a35cfbeb136465f : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":3},{"id":"37f308189699bc6d8a9fb040e71964969362cdc0","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-19 12:42:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver build/fe04f8157342427d9ba64c87992ada7a : FAILURE in 34m 55s","accounts_in_message":[],"_revision_number":3},{"id":"ef2726513fb38550f5978395b28a5a2eff51d008","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-19 12:46:00.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic6afb506ba4727746925368dc714d829967be3a7 .","accounts_in_message":[],"_revision_number":3},{"id":"ade66763726a581639275f3110864827a0eb82e0","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-19 12:46:51.000000000","message":"Patch Set 3:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/964289 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"2ec6b8a4733ff4f88ac09a34d961036b5163d935","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-19 12:57:10.000000000","message":"Patch Set 3:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/964289 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"2dafac38decd8e684abe74bf8684f967549b2739","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-19 12:58:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/964289 : FAILURE in 40m 00s","accounts_in_message":[],"_revision_number":3},{"id":"86f7f89fc43108a3a2e2a3e1abf8d32ee6de368f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-19 12:58:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/00e40b25591e44db8b1d84536fd367b0 : FAILURE in 40m 19s","accounts_in_message":[],"_revision_number":3},{"id":"826efcdb2f0fda5c55d9fdf14c62867c851cb7cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 13:03:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ac2896fcb747aaa2845309871f9fe8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c0786d7193484e059be8ce1a464f1b1c : POST_FAILURE in 13m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/afb7337b65794b1fad86fd37fef852d1 : SUCCESS in 14m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3110b9d6efc04e1cad2acf4efe6df8e3 : RETRY_LIMIT in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7644d71d721b5ec8628e2396ad3c59b256beafdc","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-19 13:08:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/96bb752400234e5cb3fdf666685e00fc : FAILURE in 45m 54s","accounts_in_message":[],"_revision_number":3},{"id":"75c7c71f66a8cba0d3e620f1b7a46143ba855a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 16:02:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d78c573257c4fccb21bdb5cf7b1b4c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62be030979694744876f01732da9db6d : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a01f341cbd64422960bfd6fab45a6aa : POST_FAILURE in 9m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec73a4b51942498a9e02027b6c219e11 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ac3fc79d0a4f8d88d8ce414f6aaef2 : SUCCESS in 15m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/9d2c41ee07d8405380501b7b0944610b : POST_FAILURE in 1h 02m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f4368944d771405a96d561fc7d144942 : POST_FAILURE in 1h 17m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/914a956e0b2c4443bd6174820d6aac1e : SUCCESS in 1h 36m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/268c9cf76e1c4b30b9cb7548db7c03d2 : SUCCESS in 1h 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb0f86b9efc04d2f965fb4fa1a40ae42 : SUCCESS in 16m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9ddf39c221f64cd7bd88f0150721b934 : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8aa784c841434f75a4af8185f2f649fe : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/127c7266d5f8447286f30e48ec058ab7 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9039c34259cc405ca9e157999c3a3ae0 : POST_FAILURE in 5m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7cef610e24ce434983d8aae6b9166881 : SUCCESS in 4m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a39b95d83d9c4746a80cf7105207f13a : FAILURE in 45m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/31117cb2236b4b42a2e05d3e6585196d : FAILURE in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/462255ac8c0641bdb50071d0ba46f262 : SUCCESS in 1h 50m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/478d0c4e286f45e8bda2067f4b17086a : SUCCESS in 2h 18m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4677b07ba4fe4717a688213d0b88cc1b : SUCCESS in 1h 49m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ad0f7ecec0e24d3dbc05736c2931b4d7 : FAILURE in 1h 33m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eab12c0ee44c4d56b8719ab2796d64bb : SUCCESS in 37m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5c0dd9321b7f4161818545a2777bbf63 : POST_FAILURE in 1h 01m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/491d9c6020e943f482e74ce3bfcf7a66 : SUCCESS in 1h 34m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b054e8096fc84a0d96af40d43e870f3f : SUCCESS in 47m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fa15eea0b19b4c6d8f25ef4d004dd3e8 : SUCCESS in 2h 07m 07s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/cb84ac01c3514516a6ababcfd7d99cb5 : SUCCESS in 1h 10m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/431e987450334649b55818e4c77261bf : SUCCESS in 1h 41m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddad2bbb8d6f43848aa74690b5ac3124 : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":3},{"id":"28063de7f4430feba297bb0debb1a2d96e35dda6","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-19 20:09:21.000000000","message":"Patch Set 3:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_761/index.html : SUCCESS in 87m","accounts_in_message":[],"_revision_number":3},{"id":"80035e300be347b1cf2847023d8e9656d824036d","author":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"date":"2026-02-23 09:34:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca6d7250748bb744d102d33688822eb77fd51973","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-23 09:39:54.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f419908daa6144069d111c02ab842ff8 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f3c48519158948c29998dc2ba4642ac8 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":3},{"id":"def31c26d967b29ffe3b647378de632a206612b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 09:55:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9b4fc0186674198921d861cc0536831\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/224ff9fd5e034cb380a4b011929fd0a2 : SUCCESS in 13m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f31ad9508f743a7980d35f024f4f1b6 : SUCCESS in 15m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9903b717e4d34a30b90179ceca82109c : RETRY_LIMIT in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9dc8069d8e9fe3d3b506aadad30c4a25a64dbe80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 12:49:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95d870177c864d2cb31ec71a61079608\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c045be324e44e5ca7ba978381e1ab93 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ca517cd7bcb45feb337f439f1da9276 : SUCCESS in 8m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/846143aebf0f438dbe8e0cced58efe36 : SUCCESS in 7m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38a26c42b5d94842807e46c8bb20bfb2 : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33bca876ad9d4311b4dee48e39a96513 : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/b6e0687dfe21433b91d8087b3f9b6822 : SUCCESS in 58m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7eae47d5229f444fbffe327d5f18e438 : SUCCESS in 31m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a30fbc10ba444918163e7374be49345 : SUCCESS in 1h 28m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbb0a627aa714e20b99d3052d2f11970 : SUCCESS in 51m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86a08632e52f40ccbd5d1b3d2da9e7f1 : SUCCESS in 14m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0cfce06dfd4c4310967b4e51500ae93b : SUCCESS in 15m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f7bdf50529544cf0b4b28aecd8fe0cf3 : SUCCESS in 2m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2aa3af7b8c6943049ea07a0d259adab5 : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15dbc8752dcf49919d0391d63dc46185 : SUCCESS in 6m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a1e26179172c446c877dca9b62ae0d79 : SUCCESS in 5m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a316e8e15e8b4e7b9c73e3323be2da08 : FAILURE in 21m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/80328d8d16fd436fad6376ed50993760 : FAILURE in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/078ba88717cc4d6d9a0f7da597de8a65 : SUCCESS in 1h 51m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0904f271b2424d5b8a95991b60f04fe5 : FAILURE in 2h 29m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/47f3b63d46754cf1ae1adb60515ab95b : SUCCESS in 1h 40m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fd51ded1690c4f1e814da53a52815456 : SUCCESS in 1h 29m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3087ae3e34444911bf627b10ae415a15 : SUCCESS in 19m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c5d9a9e8c0d344b9ac8a82c999c2dd2d : SUCCESS in 40m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/27546a804a5d4338bfddf39ff16f6787 : SUCCESS in 1h 32m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/81d8f04b42c44c459504f3c8f66d7aca : SUCCESS in 24m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0a13139fdf21429bb2466087a86c340c : SUCCESS in 1h 09m 00s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/7e53ea33b6a747aa9bdb58e44ced03ea : TIMED_OUT in 3h 11m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b35b690480b441bf969b79d5c585141f : SUCCESS in 1h 54m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8178262b884b4458ad114c802ba0cffb : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":3},{"id":"30f8ed23130e28a2d1ca1d42c1e1f8c25bd7b327","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-23 13:04:43.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/23_02_2026_12_52_45\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"c2d277eb423ac025a6195a2f68a3a560241dd53f","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2026-02-23 14:15:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a42e9d6abd3687189d85a7ba2cddc52d91766256","author":{"_account_id":35467,"name":"Luan Utimura","display_name":"Luan Utimura","email":"luan.utimura@luizalabs.com","username":"lutimura"},"date":"2026-04-02 16:41:27.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"04f0cf598bd95ebc2b97b93470700e9faee03c4c","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-06-05 14:47:34.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4cf7cbca5fca499bac332966018755a2893d67ef","author":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"date":"2026-08-26 08:56:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"59343539c2741d0710312474fecbf01082a4bbee","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-02 15:23:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d18ae5f94ea9c7c22e95415bcd45647ce89e413f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"date":"2026-09-04 07:44:59.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6462d1d0a61c40611b363b94d2b62101c2a6c93a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"date":"2026-09-04 07:54:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b44300f587a005bd6157cc28d4c1c98f25ce8e29","author":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"date":"2026-09-04 07:56:00.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a66c1d1ca053899c53926d568a1b1c3640103a63","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-04 08:28:10.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest build/7dc96c1134704402a35abce0edf4416e : RETRY_LIMIT in 31m 56s\n- cinder-storpool-tempest-iscsi-multipath build/4098da8a0c3b4731aec843532bab6bd8 : RETRY_LIMIT in 32m 00s","accounts_in_message":[],"_revision_number":5},{"id":"7bfd8020192f036a08febfdc72eb0b1c71317242","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-04 08:29:10.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/e86edbdc07e2474fac9cb351dd474ee4/","accounts_in_message":[],"_revision_number":5},{"id":"fd90bb55d26004bd442e508bf1bfb98682ce1420","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 08:41:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5407/.","accounts_in_message":[],"_revision_number":4},{"id":"29b7b5a221b5300e309ffeb3f77c33de23143ca7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 08:41:43.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic6afb506ba4727746925368dc714d829967be3a7 .","accounts_in_message":[],"_revision_number":4},{"id":"7fab5e2d19e2fdb5f3bac134950aa9e313a1a524","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-04 08:47:28.000000000","message":"Patch Set 5:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1441 : SUCCESS in 0h 49m 55s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/964289","accounts_in_message":[],"_revision_number":5},{"id":"1933ec820b943c24a96cc27febf64c19988f7cec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-04 08:48:04.000000000","message":"Patch Set 5:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1173 : SUCCESS in 0h 50m 29s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/964289","accounts_in_message":[],"_revision_number":5},{"id":"57959e3e1d79871d7196acb49f286f01f595badd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 08:48:44.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic6afb506ba4727746925368dc714d829967be3a7 .","accounts_in_message":[],"_revision_number":5},{"id":"5b01218315fc4ef24ec06d10f2b9c5548c03e658","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-04 08:48:48.000000000","message":"Patch Set 5:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1080 : SUCCESS in 0h 51m 19s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/964289","accounts_in_message":[],"_revision_number":5},{"id":"d173cf98f04ce2958c57af443030e38a28af2cad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 08:58:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bca05844f1d4668865ffe05062abeae\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b43f252904b84e12bfd27b21856cc1a9 : SUCCESS in 19m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4b2cf7f1aae2482b85c9c1f596e895fc : SUCCESS in 20m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e4816b7b4cbd41ba9870d8461b09a5f0 : SUCCESS in 26m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b8db17ec46438cf99070fa9771eb7e1a2c714cfc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-04 09:17:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e86edbdc07e2474fac9cb351dd474ee4\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a5a4296135984dbdadb1bd963c85a240 : SUCCESS in 48m 05s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ec1682ba9ba148b09f2c9876b45b0e13 : SUCCESS in 36m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a1e94598656c40b28b5d5a8960bfa050 : SUCCESS in 1h 01m 42s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a7b050bad56f451f974fdba233945426 : SUCCESS in 46m 44s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/82492fa662ba4567aaa4ba16d7e87866 : FAILURE in 31m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"41aa9a0b150a3767e0f0047965e92b3b645e0c9e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 09:37:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5408/.","accounts_in_message":[],"_revision_number":5},{"id":"9786e8fee975f1c4140cdc1987b1a337c3cfaf2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 10:13:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb69e52229d3499e8a8adef6242e19e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6fade24a1d5471ba98cc8f47b32150d : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e842792b8e44c8e82bc864430bc9156 : SUCCESS in 13m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63a84aad1d534bcbbc63119f5e4d7dc1 : SUCCESS in 6m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c86533557c054e7faa557fecc08f21cd : SUCCESS in 14m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f5bbde59af4aa8aa78dcec2d38dd33 : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/b491f1d71db34dd8af81d90f33b31197 : SUCCESS in 1h 05m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/681e6549070f4a51997bf20a4a75fa0f : SUCCESS in 30m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e0ecf270971496abe1005aabb393478 : SUCCESS in 1h 08m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1bd40bc771a4439944c8c911c032803 : SUCCESS in 1h 04m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24df288e9f3d4e85bea07943e92b452e : SUCCESS in 10m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e675873e73a1467998594dacbb7ec7e4 : SUCCESS in 21m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/07142b13f6574e44b651c00db7e8bde4 : SUCCESS in 2m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2b8bbe59674941be9558205b3330d27a : SUCCESS in 4m 19s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/0c4b6bf18aa14f398038cae1ba92ff52 : SUCCESS in 7m 23s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/91b0b4fb7c2e4001aef73ae7afe63a6f : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/59bf7bb2f0ea43ad9243cbd5020845ad : SUCCESS in 5m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/20426628129a48739a602f0af1b82671 : SUCCESS in 4m 50s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/53d65e64840e48e7b52efbe8d46eac54 : FAILURE in 5m 28s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8c9eee348149410e98489a47d0b467df : SUCCESS in 6m 11s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/253aefad305e4232a2e773d2c0272147 : SUCCESS in 36m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/08af8ae1478a466aa50bd3a978e157d5 : FAILURE in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0c379dfd096645b6b00ff7ac3d63600b : SUCCESS in 1h 49m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d54ee7b42afa4d079104670185539a24 : FAILURE in 1h 48m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f074f6c388754dac89b367972f6925cb : SUCCESS in 1h 48m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/4dcd02f32c6347d9a5ce939f1a0e156c : FAILURE in 2h 13m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/04542523f9674f2b898f135179432c74 : SUCCESS in 1h 18m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/431d241b38e944b18b913beb6c59205a : SUCCESS in 24m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd197d492b7c4bc896f1a93bd04173ac : SUCCESS in 1h 08m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/13e7d55f94234017bf32e83d3c83c1b7 : SUCCESS in 1h 06m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6159cf9b4baa48f3b66114f8baf16bf6 : SUCCESS in 41m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87f27440c72948a2a0fea848225753f8 : SUCCESS in 1h 28m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8468e95045004f9b9c1f0668e28d05ba : SUCCESS in 26m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"59ec1b6c25a57cfa6c5f1c3b9cf360c121c9c71b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-04 10:22:55.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_09_2026_10_16_47\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"f344d435d3746fffd8c6d4d4af9f69323c94489e","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-04 10:24:44.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-89-964289-5-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-89-964289-5-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d0e370730ee37a0255b5d5e6592957106c3c2977","revisions":{"5b8e34b268783f81afbc3fbdbbb0280448f421d5":{"kind":"REWORK","_number":1,"created":"2025-10-17 08:50:41.000000000","uploader":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"ref":"refs/changes/89/964289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/964289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/964289/1"}}},"commit":{"parents":[{"commit":"eb3666620dc455a39f005b95d7f286ba17896a9f","subject":"Merge \"NetApp - Add additional debug logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb3666620dc455a39f005b95d7f286ba17896a9f"}]}],"author":{"name":"Moritz Wanzenböck","email":"moritz.wanzenboeck@linbit.com","date":"2025-10-17 08:28:52.000000000","tz":120},"committer":{"name":"Moritz Wanzenböck","email":"moritz.wanzenboeck@linbit.com","date":"2025-10-17 08:50:30.000000000","tz":120},"subject":"[LINSTOR] rewrite driver from scratch","message":"[LINSTOR] rewrite driver from scratch\n\nThis is a complete rewrite of the LINSTOR driver, using the existing API\nbindings form the python-linstor project. The replaced driver had a number\nof design decisions that turned out to be mistakes:\n\n- Automatic configuration of LINSTOR storage pools. This barely worked in\n  test set up, and is best left to the administrator using the \"linstor\"\n  CLI command.\n- There was an option to \"downsize\" volumes about a fixed size, which does\n  not make sense: if a 10GiB volume is requested, we should create a 10GIB\n  volume.\n- There was code that manually deployed replicas of a volume on every node.\n  This can be handled in LINSTOR directly without the need for additional\n  code in this driver.\n- There was some custom handling of block_size settings and other \"target\"\n  related settings. This is best left to the regular target helpers.\n- The user had to manually ensure that the cinder volume service host has\n  access to the volume. This should generally be managed dynamically by the\n  driver.\n\nAs such, the replacement driver fixes the above issues. Care has been taken\nthat existing volumes continue to work after the update. In addition to\nreimplementing the existing features (where it makes sense), the following\nfeatures have been added:\n\n- Specifying multiple LINSTOR Controller URLs, which simplifies HA setups with\n  additional load balancers.\n- The LINSTOR concept of \"Resource Groups\" is now used to implement Cinder\n  Volume Types, letting users dynamically set replica counts, DRBD settings,\n  etc...\n- Volume backed live migration for directly attached DRBD volumes\n- Multi-Attach for iSCSI exported volumes\n- Snapshot revert support\n- Assisted retype and volume type migration\n\nThe unit tests have been updated to conform with the python-linstor library.\n\nChange-Id: Ic6afb506ba4727746925368dc714d829967be3a7\nSigned-off-by: Moritz Wanzenböck \u003cmoritz.wanzenboeck@linbit.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b8e34b268783f81afbc3fbdbbb0280448f421d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b8e34b268783f81afbc3fbdbbb0280448f421d5"}]},"branch":"refs/heads/master"},"49b77cbadd8b9aeee0906c14544abbd2783c07af":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-05 12:27:57.000000000","uploader":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"ref":"refs/changes/89/964289/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/964289/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/964289/2"}}},"commit":{"parents":[{"commit":"c9f569a4e434bef7f3a7a442ddaeba6621100732","subject":"Merge \"[NetApp] Default node model to empty string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9f569a4e434bef7f3a7a442ddaeba6621100732"}]}],"author":{"name":"Moritz Wanzenböck","email":"moritz.wanzenboeck@linbit.com","date":"2025-10-17 08:28:52.000000000","tz":120},"committer":{"name":"Moritz Wanzenböck","email":"moritz.wanzenboeck@linbit.com","date":"2025-11-05 12:27:41.000000000","tz":60},"subject":"[LINSTOR] rewrite driver from scratch","message":"[LINSTOR] rewrite driver from scratch\n\nThis is a complete rewrite of the LINSTOR driver, using the existing API\nbindings form the python-linstor project. The replaced driver had a number\nof design decisions that turned out to be mistakes:\n\n- Automatic configuration of LINSTOR storage pools. This barely worked in\n  test set up, and is best left to the administrator using the \"linstor\"\n  CLI command.\n- There was an option to \"downsize\" volumes about a fixed size, which does\n  not make sense: if a 10GiB volume is requested, we should create a 10GIB\n  volume.\n- There was code that manually deployed replicas of a volume on every node.\n  This can be handled in LINSTOR directly without the need for additional\n  code in this driver.\n- There was some custom handling of block_size settings and other \"target\"\n  related settings. This is best left to the regular target helpers.\n- The user had to manually ensure that the cinder volume service host has\n  access to the volume. This should generally be managed dynamically by the\n  driver.\n\nAs such, the replacement driver fixes the above issues. Care has been taken\nthat existing volumes continue to work after the update. In addition to\nreimplementing the existing features (where it makes sense), the following\nfeatures have been added:\n\n- Specifying multiple LINSTOR Controller URLs, which simplifies HA setups with\n  additional load balancers.\n- The LINSTOR concept of \"Resource Groups\" is now used to implement Cinder\n  Volume Types, letting users dynamically set replica counts, DRBD settings,\n  etc...\n- Volume backed live migration for directly attached DRBD volumes\n- Multi-Attach for iSCSI exported volumes\n- Snapshot revert support\n- Assisted retype and volume type migration\n\nThe unit tests have been updated to conform with the python-linstor library.\n\nChange-Id: Ic6afb506ba4727746925368dc714d829967be3a7\nSigned-off-by: Moritz Wanzenböck \u003cmoritz.wanzenboeck@linbit.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49b77cbadd8b9aeee0906c14544abbd2783c07af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49b77cbadd8b9aeee0906c14544abbd2783c07af"}]},"branch":"refs/heads/master"},"3c36dae62bee66eff93ed30f774b035fa7bea607":{"kind":"REWORK","_number":3,"created":"2026-02-19 12:06:39.000000000","uploader":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"ref":"refs/changes/89/964289/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/964289/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/964289/3"}}},"commit":{"parents":[{"commit":"347b871e44d8f0556f2be3e0967294b6394febed","subject":"Merge \"Doc: remove python 2 references, no longer relevant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/347b871e44d8f0556f2be3e0967294b6394febed"}]}],"author":{"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","date":"2026-02-19 11:04:02.000000000","tz":60},"committer":{"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","date":"2026-02-19 11:49:45.000000000","tz":60},"subject":"linstor: rewrite driver v2.0.0","message":"linstor: rewrite driver v2.0.0\n\nThis is a complete rewrite of the LINSTOR driver, using the existing API\nbindings form the python-linstor project. The replaced driver had a number\nof design decisions that turned out to be mistakes:\n\n- Automatic configuration of LINSTOR storage pools. This barely worked in\n  test set up, and is best left to the administrator using the \"linstor\"\n  CLI command.\n- There was an option to \"downsize\" volumes about a fixed size, which does\n  not make sense: if a 10GiB volume is requested, we should create a 10GIB\n  volume.\n- There was code that manually deployed replicas of a volume on every node.\n  This can be handled in LINSTOR directly without the need for additional\n  code in this driver.\n- There was some custom handling of block_size settings and other \"target\"\n  related settings. This is best left to the regular target helpers.\n- The user had to manually ensure that the cinder volume service host has\n  access to the volume. This should generally be managed dynamically by the\n  driver.\n\nAs such, the replacement driver fixes the above issues. Care has been taken\nthat existing volumes continue to work after the update. In addition to\nreimplementing the existing features (where it makes sense), the following\nfeatures have been added:\n\n- Specifying multiple LINSTOR Controller URLs, which simplifies HA setups with\n  additional load balancers.\n- The LINSTOR concept of \"Resource Groups\" is now used to implement Cinder\n  Volume Types, letting users dynamically set replica counts, DRBD settings,\n  etc...\n- Volume backed live migration for directly attached DRBD volumes\n- Multi-Attach for iSCSI exported volumes\n- Snapshot revert support\n- Assisted retype and volume type migration\n\nThe unit tests have been updated to conform with the python-linstor library.\n\nAdditionally, an option to set a custom hostname was added. LINSTOR driver\nneeds to know the hostname of the cinder-volume process. However, when\ndeployed in a container environment, this host name is not always the \"real\"\nhostname. To fix this, pass in the real hostname as a file, pointing the\nLINSTOR driver at that file.\n\nDocument the complete rewrite of the LINSTOR Cinder driver, including\nnew features (Resource Group mapping, HA controller support, live\nmigration, multi-attach, snapshot revert, TLS), upgrade notes for\nadministrators, and the fix for the broken iSCSI export path.\n\nChange-Id: Ic6afb506ba4727746925368dc714d829967be3a7\nSigned-off-by: Moritz Wanzenböck \u003cmoritz.wanzenboeck@linbit.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c36dae62bee66eff93ed30f774b035fa7bea607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c36dae62bee66eff93ed30f774b035fa7bea607"}]},"branch":"refs/heads/master"},"fc472858745b1e6448d54814719b2f6bfdbc5d4c":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2026-09-04 07:44:59.000000000","uploader":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"ref":"refs/changes/89/964289/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/964289/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/964289/4"}}},"commit":{"parents":[{"commit":"c409487fb0e929e02e8e07e391c99d21025cf5b6","subject":"Merge \"NetApp ONTAP retype/migrate to a non-NetApp backend fails with host \u0027none\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c409487fb0e929e02e8e07e391c99d21025cf5b6"}]}],"author":{"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","date":"2026-02-19 11:04:02.000000000","tz":60},"committer":{"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","date":"2026-09-04 07:44:57.000000000","tz":120},"subject":"linstor: rewrite driver v2.0.0","message":"linstor: rewrite driver v2.0.0\n\nThis is a complete rewrite of the LINSTOR driver, using the existing API\nbindings form the python-linstor project. The replaced driver had a number\nof design decisions that turned out to be mistakes:\n\n- Automatic configuration of LINSTOR storage pools. This barely worked in\n  test set up, and is best left to the administrator using the \"linstor\"\n  CLI command.\n- There was an option to \"downsize\" volumes about a fixed size, which does\n  not make sense: if a 10GiB volume is requested, we should create a 10GIB\n  volume.\n- There was code that manually deployed replicas of a volume on every node.\n  This can be handled in LINSTOR directly without the need for additional\n  code in this driver.\n- There was some custom handling of block_size settings and other \"target\"\n  related settings. This is best left to the regular target helpers.\n- The user had to manually ensure that the cinder volume service host has\n  access to the volume. This should generally be managed dynamically by the\n  driver.\n\nAs such, the replacement driver fixes the above issues. Care has been taken\nthat existing volumes continue to work after the update. In addition to\nreimplementing the existing features (where it makes sense), the following\nfeatures have been added:\n\n- Specifying multiple LINSTOR Controller URLs, which simplifies HA setups with\n  additional load balancers.\n- The LINSTOR concept of \"Resource Groups\" is now used to implement Cinder\n  Volume Types, letting users dynamically set replica counts, DRBD settings,\n  etc...\n- Volume backed live migration for directly attached DRBD volumes\n- Multi-Attach for iSCSI exported volumes\n- Snapshot revert support\n- Assisted retype and volume type migration\n\nThe unit tests have been updated to conform with the python-linstor library.\n\nAdditionally, an option to set a custom hostname was added. LINSTOR driver\nneeds to know the hostname of the cinder-volume process. However, when\ndeployed in a container environment, this host name is not always the \"real\"\nhostname. To fix this, pass in the real hostname as a file, pointing the\nLINSTOR driver at that file.\n\nDocument the complete rewrite of the LINSTOR Cinder driver, including\nnew features (Resource Group mapping, HA controller support, live\nmigration, multi-attach, snapshot revert, TLS), upgrade notes for\nadministrators, and the fix for the broken iSCSI export path.\n\nChange-Id: Ic6afb506ba4727746925368dc714d829967be3a7\nSigned-off-by: Moritz Tanner \u003cmoritz.wanzenboeck@linbit.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc472858745b1e6448d54814719b2f6bfdbc5d4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc472858745b1e6448d54814719b2f6bfdbc5d4c"}]},"branch":"refs/heads/master"},"d0e370730ee37a0255b5d5e6592957106c3c2977":{"kind":"REWORK","_number":5,"created":"2026-09-04 07:54:10.000000000","uploader":{"_account_id":33067,"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","username":"wanzenbug"},"ref":"refs/changes/89/964289/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/964289/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/964289/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/964289/5"}}},"commit":{"parents":[{"commit":"c409487fb0e929e02e8e07e391c99d21025cf5b6","subject":"Merge \"NetApp ONTAP retype/migrate to a non-NetApp backend fails with host \u0027none\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c409487fb0e929e02e8e07e391c99d21025cf5b6"}]}],"author":{"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","date":"2026-02-19 11:04:02.000000000","tz":60},"committer":{"name":"Moritz Tanner","email":"moritz.wanzenboeck@linbit.com","date":"2026-09-04 07:54:07.000000000","tz":120},"subject":"linstor: rewrite driver v2.0.0","message":"linstor: rewrite driver v2.0.0\n\nThis is a complete rewrite of the LINSTOR driver, using the existing API\nbindings form the python-linstor project. The replaced driver had a number\nof design decisions that turned out to be mistakes:\n\n- Automatic configuration of LINSTOR storage pools. This barely worked in\n  test set up, and is best left to the administrator using the \"linstor\"\n  CLI command.\n- There was an option to \"downsize\" volumes about a fixed size, which does\n  not make sense: if a 10GiB volume is requested, we should create a 10GIB\n  volume.\n- There was code that manually deployed replicas of a volume on every node.\n  This can be handled in LINSTOR directly without the need for additional\n  code in this driver.\n- There was some custom handling of block_size settings and other \"target\"\n  related settings. This is best left to the regular target helpers.\n- The user had to manually ensure that the cinder volume service host has\n  access to the volume. This should generally be managed dynamically by the\n  driver.\n\nAs such, the replacement driver fixes the above issues. Care has been taken\nthat existing volumes continue to work after the update. In addition to\nreimplementing the existing features (where it makes sense), the following\nfeatures have been added:\n\n- Specifying multiple LINSTOR Controller URLs, which simplifies HA setups with\n  additional load balancers.\n- The LINSTOR concept of \"Resource Groups\" is now used to implement Cinder\n  Volume Types, letting users dynamically set replica counts, DRBD settings,\n  etc...\n- Volume backed live migration for directly attached DRBD volumes\n- Multi-Attach for iSCSI exported volumes\n- Snapshot revert support\n- Assisted retype and volume type migration\n\nThe unit tests have been updated to conform with the python-linstor library.\n\nAdditionally, an option to set a custom hostname was added. LINSTOR driver\nneeds to know the hostname of the cinder-volume process. However, when\ndeployed in a container environment, this host name is not always the \"real\"\nhostname. To fix this, pass in the real hostname as a file, pointing the\nLINSTOR driver at that file.\n\nDocument the complete rewrite of the LINSTOR Cinder driver, including\nnew features (Resource Group mapping, HA controller support, live\nmigration, multi-attach, snapshot revert, TLS), upgrade notes for\nadministrators, and the fix for the broken iSCSI export path.\n\nChange-Id: Ic6afb506ba4727746925368dc714d829967be3a7\nSigned-off-by: Moritz Tanner \u003cmoritz.wanzenboeck@linbit.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d0e370730ee37a0255b5d5e6592957106c3c2977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d0e370730ee37a0255b5d5e6592957106c3c2977"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
