)]}'
{"id":"openstack%2Fcinder~964454","triplet_id":"openstack%2Fcinder~stable%2F2024.2~I0c6b2c584e8e0053ad740ee25f29ca1d442bdeea","project":"openstack/cinder","branch":"stable/2024.2","topic":"bug/1886543","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-21 18:15:25.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"38329":{"account":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"last_update":"2025-10-21 12:30:08.000000000","reason":"removed on reply"},"38493":{"account":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"last_update":"2025-10-22 23:28:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c6b2c584e8e0053ad740ee25f29ca1d442bdeea","subject":"Driver assisted migration on retype when it\u0027s safe","status":"MERGED","created":"2025-10-21 09:32:26.000000000","updated":"2025-10-22 23:29:43.000000000","submitted":"2025-10-22 23:28:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964454-bug/1886543","meta_rev_id":"ea485e1ec33e1c6267a0f56b016dc101516e8b69","_number":964454,"virtual_id_number":964454,"owner":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-22 23:28:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-22 21:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-22 12:44:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-10-21 18:15:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-10-21 12:30:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-10-22 21:38:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-21 09:32:26.000000000","updated_by":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"CC"},{"updated":"2025-10-21 09:33:16.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-21 09:39:01.000000000","updated_by":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"reviewer":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"state":"REVIEWER"},{"updated":"2025-10-21 09:39:01.000000000","updated_by":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-21 09:54:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-21 09:55:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-10-21 10:10:11.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-10-21 12:40:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-21 20:29:14.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-10-22 12:44:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-10-22 21:38:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"601cd7792dc589452b063947e69d0368160008a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"date":"2025-10-21 09:32:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6dc33ac04a315f78488272e2725f9b785234c1d6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-21 09:33:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"7ec597dc705af33780e7baf145ee748b3a0a9833","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 09:54:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25ad92d021ea4d56ab3f95709b045887\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/14f1adc8c1f448aa935b80333b2b487e : SUCCESS in 20m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c31baf3051594b65b1eb06953f75ce24 : SUCCESS in 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ad37e8aad39fc095bcd62bf898b1220c998a40d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-21 09:55:17.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2201/.","accounts_in_message":[],"_revision_number":1},{"id":"405788f4ef581abd225e65a4bfa643b40ecd00dd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-21 09:55:17.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2087/.","accounts_in_message":[],"_revision_number":1},{"id":"7675497c33bfdfc6978539b4a6b261934d0835f1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-21 10:10:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/964454 : FAILURE in 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7066d437b2cfd12fff9cd6ed696e9c4d99e7b8e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"date":"2025-10-21 10:34:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"96b374327fb24721bf4086d27d8a92c9f71d9d52","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-21 10:35:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"a6419b63d93852d01c195876378cdb4cd3ea1afb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 10:56:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4da78789c72a46069d497f477df4c9d4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54a3d1dea65641d5b4a6b3c5ee6f65e9 : SUCCESS in 18m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d6e79a5427b84c52a731bdc61475c551 : SUCCESS in 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89831e0b52389b325d5831643d9fddac36f350ff","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-21 11:47:17.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2090/.","accounts_in_message":[],"_revision_number":2},{"id":"e2f55fc9f30ce3aeda635f25be501d2906df5fef","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-21 11:48:28.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2204/.","accounts_in_message":[],"_revision_number":2},{"id":"391e15a9b76da0de95ae9a24de42c96dddb68541","author":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"date":"2025-10-21 12:30:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a25b609ebae7e6a1cd4cef29893adaf6686fd4c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 12:40:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da5f6d929aa74d0b86e6f458a608cd5d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/235b1b3abfca4e1f87d3cf9799491332 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc3e739b03fb4b78a8f9783f4d6eb7de : SUCCESS in 11m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e6cf8fca5404ca38c41178dc7f0f489 : SUCCESS in 11m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9a2d9f76bea433ebd21879029f3da86 : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21cbfbde4ff84f6caf0cba0e64bb3638 : SUCCESS in 12m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/614e3dd2572f4d1cbf1e65bd5c4448ce : SUCCESS in 39m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1ef1a2e9a56747fdb4ec46016c7c995e : SUCCESS in 1h 38m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac4893c57e7142e4b838cf28ab832ba6 : FAILURE in 33m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2954b7d570d94bbbb640eff0becbf551 : SUCCESS in 15m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8f523c966002416b996bb014e841ce54 : SUCCESS in 17m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8e09d8b56226472f8a7a0210035920f8 : SUCCESS in 3m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/59586f7bcca34f618ea6cad562b52c01 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/238ad2908a244a6fb27e97f1cd9718bb : SUCCESS in 5m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/190c75a7b830432aacb62da46e102153 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/341fa10ad1cb44f9b48654749897a356 : FAILURE in 32m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b445412a58548d0b5d99452292ae3f9 : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/38032964582849a6bd7082b4b5543c51 : SUCCESS in 1h 09m 05s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e018da6e69034932b7068d2f0e0cc387 : SUCCESS in 1h 20m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8acf8a78adf7413fb91e8a1db6406abb : SUCCESS in 1h 57m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/701342f64f45400994363899faf30926 : SUCCESS in 27m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/46b1317ee5d9482baf4a131ed1755ef7 : SUCCESS in 1h 09m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/674d9f5f7b204e42951283ef032c4fd9 : SUCCESS in 1h 21m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aecbd4b52ee94c4ab3289e03b22ea6f1 : SUCCESS in 42m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/165cc1c30f5f4e3b9c63d0c68ebda7e6 : SUCCESS in 1h 11m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/423efa7a65074b5b886dd124c4171a19 : SUCCESS in 2h 03m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0b1deb148424e67abbabadf2184c370 : SUCCESS in 28m 56s","accounts_in_message":[],"_revision_number":2},{"id":"785032842a5937249005d53d0b25c22b67fc94dc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-21 13:21:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/964454 : FAILURE in 36m 25s","accounts_in_message":[],"_revision_number":2},{"id":"e3fe1005ca7455cf09bc9c8e1c00a3810180aa06","author":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"date":"2025-10-21 13:32:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d14302b4df342321ad9628c82774a7c0a7c5bd76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 14:00:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10c87a74de684cc98c770e0b102b0d17\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9da2314df5a4198a4bb2b7839a4c31c : SUCCESS in 26m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3554c44a358e458790035068f2d9a9f2 : SUCCESS in 24m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09f1c6d022bdc7c532f9a4b270c377132c0e1b35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 15:38:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2383124a7aef43ad87ca8497f1df918b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d3e75332234423294daad8b76ab0965 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54422122563b45f18fa40f8a1175bcb0 : SUCCESS in 7m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ae49d531dd34d5db55f802a373ba0b5 : SUCCESS in 7m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/246296e2d7ac4a178bc42e573b318565 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5cca23a61e64c74bf5732823f682d89 : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/0be1b1141fb5497ea6e59404f673fd63 : SUCCESS in 27m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b6917d789ac4f93a2ff2e67521371dc : SUCCESS in 1h 26m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb5c8367ccb34b6f972d5bf49a391676 : SUCCESS in 34m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e81ad90464fb4749896c77d72e0da73a : SUCCESS in 19m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c831db6085a0415c9625877ca91cec18 : SUCCESS in 11m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a71a52e5b2a341388e577cda0425197d : POST_FAILURE in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/869a76a5a2874b60bb17b843cb1218ea : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9de17dd6274e470db266a96e66ebee78 : SUCCESS in 9m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cdf39a318167457e9946a817e27b967b : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7b59c6854f764b6ca672b6b5c6f80b87 : FAILURE in 36m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/96a32a1b973449088d1aef984506f98e : SUCCESS in 6m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/812e677d5fb94322bef8767b5b2b25c3 : SUCCESS in 1h 50m 02s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/523c0bdd12c844239b2245702a35a2d2 : SUCCESS in 1h 31m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/76a7a0cad7d44919b372226581c32c9b : SUCCESS in 1h 53m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/764a33009a5a43319c75e53d7db3777f : SUCCESS in 40m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/54da48251a1d4e97aa66a4f286b9db85 : SUCCESS in 50m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/438008e60bcf409d8d4ac7944ec23a5b : SUCCESS in 1h 12m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d00247c44ed04ff5bbabf5dd6fe1fa34 : SUCCESS in 35m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/56daea24e976420f9136afedaccd7c1e : SUCCESS in 2h 00m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1c51882aaa2f4557b940102f443a2829 : SUCCESS in 1h 07m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b10db2705ff46a688d8cd621dd82a87 : SUCCESS in 35m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b9c2a6923ad4e180dca622bcc0f26be1bb1a59b6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-21 18:15:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed9df89e9bb06abea8af722054c897953bdb30ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 18:36:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb01e10838ff4783aa53820c29ae69e7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1ccd4fbbdc647d48d448fd0c493370c : SUCCESS in 19m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a60cbbba4259429b9a3e7871e4466f34 : SUCCESS in 17m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d69807139cadf16945dd21e8ba630bf3bd3a4842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 20:22:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b64b399458844892a1bc0fe5c4c67e7a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1325dff93df46d1a42c3d8033442033 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7ccb17b1c4046fb92308b93787a8fdd : SUCCESS in 10m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eca4613071d54b54995b1d926f9ae907 : SUCCESS in 11m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f47b1655644b4026b1f2904b85a03f23 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9656f34b873a45eab9807a7c47e4a9f1 : SUCCESS in 12m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/a5dc62cbe60a47d0b2881a38b562ce91 : SUCCESS in 31m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90bb860b72c94d0eb2408c34eb9216ad : SUCCESS in 1h 20m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a17e9f48ee14644831f3f787d84f978 : SUCCESS in 57m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02b2612d5d644b0e95cff44c25801179 : SUCCESS in 15m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ef836060bf3b46869c8a37cc4e2fdf8d : SUCCESS in 11m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a34fdf1878a447d28df560b0c5452443 : SUCCESS in 2m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/773a322a22404a678574c2934f665714 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/03ede597dabd4f94bd3a5cdd62220eee : SUCCESS in 5m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/01c266538e5f46b5ab4e83b33fb02764 : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b2d2dc78d5174f1a940fa680dacad0b6 : FAILURE in 34m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eeaf83a0714f48bbb78286601c7ebfdb : SUCCESS in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/50bcbd8d83314361a6baaa4f99842edb : SUCCESS in 1h 40m 50s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/85fa06ee2d1b41a891c7ba2bb00d41ec : SUCCESS in 1h 36m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/df6e3ced8b0d48618c55a075c14146a2 : SUCCESS in 1h 22m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/989e07069234496492821f78e3192406 : SUCCESS in 31m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/34cf2ebf8b8d42be93406d2ea2ff6ebe : SUCCESS in 1h 04m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c7ce56297ccf44648e7077f5188c6e29 : SUCCESS in 1h 20m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dfd6f82b896845d2ab0ec582d866782e : SUCCESS in 39m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7ae644209d7945419c2c211c466ce16c : SUCCESS in 1h 38m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2297ab3c5658494c8b0c74834d1ae9e2 : SUCCESS in 2h 01m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62b5c584bdf9442eb2b8fa925a7b04d2 : SUCCESS in 30m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a2aa23476efef621f89981ab4a1870b60f17879d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-21 20:29:14.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_10_2025_20_22_44\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"88e248972e4f380602c0e4e6425aefa08c45d2fe","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-22 12:44:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"753022bab7c4389c8b691e52f9e642b21aa5aba0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-22 21:38:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd570ee7ddb4f2f67a876e8d91a2c55b8b93e519","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 21:38:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e87be948f0c40c9939515b727121e76ef5538e80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 23:28:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2604560adb448bf99e7bd9561a4d53d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea1e7932bac4d60baae14d4dcba3f95 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22ad33b3456a45fa88f2839b0d7814ad : SUCCESS in 7m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1038004b9eca4b41a10717abe23ddf9a : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86da7001bbcb40aab30ecbcbfbec739b : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/f60c4fd9439b4ac39c01e0e66a400c11 : SUCCESS in 49m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59d17c8d0bf749d8a87e4f48f4ca1369 : SUCCESS in 1h 25m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02efb7f0c589419cb54eae9f12c44513 : SUCCESS in 47m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/558ec99703e8437f99be0713df2dcf4e : SUCCESS in 15m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/607ee183429d41449b6d7b455d76fec1 : SUCCESS in 38m 36s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0a1f5432d1d04644a96d482596615c1c : SUCCESS in 1h 47m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8bae08b4910a47e0a3cc75e1515f2124 : SUCCESS in 44m 35s","accounts_in_message":[],"_revision_number":2},{"id":"fca3bfe11f1a7a8426d81f62cd17f9bc3c770201","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 23:28:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ea485e1ec33e1c6267a0f56b016dc101516e8b69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 23:29:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f20ff3308a2a4a48926a39ccbaade93f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85354d61d14a45409c67ca82e6893193 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c2f0a376bca42f89d8a1907e1533f74 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49b5c010d7597dc57da81f44585133106fb8e484","revisions":{"4f18bdb5b7b25d14d73d632edb4822a282e290d7":{"kind":"REWORK","_number":1,"created":"2025-10-21 09:32:26.000000000","uploader":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"ref":"refs/changes/54/964454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/964454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/964454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/964454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/964454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/964454/1"}}},"commit":{"parents":[{"commit":"e96dcc4e9744833ccea2697c67198f55aba3ad22","subject":"HPE 3par: Ignore duplicate IP in iSCSI/vlan ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e96dcc4e9744833ccea2697c67198f55aba3ad22"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2025-01-09 17:45:27.000000000","tz":-480},"committer":{"name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","date":"2025-10-21 09:31:51.000000000","tz":300},"subject":"Driver assisted migration on retype when it\u0027s safe","message":"Driver assisted migration on retype when it\u0027s safe\n\nThis is a revision of I2532cfc9b98788a1a1e765f07d0c9f8c98bc77f6\nthat corrects the issue that forced it to be reverted by\nI893105cbd270300be9ec48b3127e66022f739314. The revised code avoids\nattempting driver assisted migration when the volume has attachments.\n\nWhen doing a retype of a volume that requires a migration, the manager\nonly uses driver assisted migration when the source and the destination\nbelong to the same backend (different pools).\n\nAs long as the volume has no attachments, driver assisted migration\nshould also be tried for other cases, just like when we do a normal\nmigration.\n\nOne case were this would be beneficial is when doing migrations from one\npool to another on the same storage system on single pool drivers (such\nas RBD/Ceph).\n\nThis patch checks what are the changes between the types to see if it is\nsafe to use driver assisted migration (from the perspective of keeping\nthe resulting volume consistent with the volume type) and when it is it\ntries to use it.\n\nIf driver assisted migration indicates that it couldn\u0027t move the volume,\nthen we go with the generic volume migration like we used to.\n\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\nCloses-Bug: #1886543\nChange-Id: I0c6b2c584e8e0053ad740ee25f29ca1d442bdeea\n(cherry picked from commit b8610a01b61cf86db2f31e26eb0d4dab4a812ace)\nSigned-off-by: Dmitrii Shevchenko \u003cdsvkrab@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f18bdb5b7b25d14d73d632edb4822a282e290d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f18bdb5b7b25d14d73d632edb4822a282e290d7"}]},"branch":"refs/heads/stable/2024.2"},"49b5c010d7597dc57da81f44585133106fb8e484":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-10-21 10:34:17.000000000","uploader":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"ref":"refs/changes/54/964454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/964454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/964454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/964454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/964454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/964454/2"}}},"commit":{"parents":[{"commit":"e96dcc4e9744833ccea2697c67198f55aba3ad22","subject":"HPE 3par: Ignore duplicate IP in iSCSI/vlan ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e96dcc4e9744833ccea2697c67198f55aba3ad22"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2025-01-09 17:45:27.000000000","tz":-480},"committer":{"name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","date":"2025-10-21 10:33:02.000000000","tz":300},"subject":"Driver assisted migration on retype when it\u0027s safe","message":"Driver assisted migration on retype when it\u0027s safe\n\nThis is a revision of I2532cfc9b98788a1a1e765f07d0c9f8c98bc77f6\nthat corrects the issue that forced it to be reverted by\nI893105cbd270300be9ec48b3127e66022f739314. The revised code avoids\nattempting driver assisted migration when the volume has attachments.\n\nWhen doing a retype of a volume that requires a migration, the manager\nonly uses driver assisted migration when the source and the destination\nbelong to the same backend (different pools).\n\nAs long as the volume has no attachments, driver assisted migration\nshould also be tried for other cases, just like when we do a normal\nmigration.\n\nOne case were this would be beneficial is when doing migrations from one\npool to another on the same storage system on single pool drivers (such\nas RBD/Ceph).\n\nThis patch checks what are the changes between the types to see if it is\nsafe to use driver assisted migration (from the perspective of keeping\nthe resulting volume consistent with the volume type) and when it is it\ntries to use it.\n\nIf driver assisted migration indicates that it couldn\u0027t move the volume,\nthen we go with the generic volume migration like we used to.\n\nCo-Authored-By: Alan Bishop \u003cabishop@redhat.com\u003e\nCloses-Bug: #1886543\nChange-Id: I0c6b2c584e8e0053ad740ee25f29ca1d442bdeea\n(cherry picked from commit b8610a01b61cf86db2f31e26eb0d4dab4a812ace)\n(cherry picked from commit 4acf0a3b6ea89c98ad3f1052c7345b0b67a5aafa)\nSigned-off-by: Dmitrii Shevchenko \u003cdsvkrab@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49b5c010d7597dc57da81f44585133106fb8e484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49b5c010d7597dc57da81f44585133106fb8e484"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
