)]}'
{"id":"openstack%2Fcinder~964492","triplet_id":"openstack%2Fcinder~master~I097408e24d8ad05aa10de26551d540c685f9c22b","project":"openstack/cinder","branch":"master","topic":"bug/2129271","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-05-05 21:22:45.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2026-05-11 17:48:04.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-05-11 17:48:04.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-05-11 17:48:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I097408e24d8ad05aa10de26551d540c685f9c22b","subject":"Properly set shared_targets for all NVMe-oF protocols","status":"MERGED","created":"2025-10-21 19:00:30.000000000","updated":"2026-06-19 20:03:50.000000000","submitted":"2026-05-11 17:48:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"964492-bug/2129271","meta_rev_id":"012a396dc9df450354c9e6e19ce6bfc20e9186b2","_number":964492,"virtual_id_number":964492,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-11 17:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-05-07 12:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-05-07 13:14:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":1,"date":"2026-05-11 16:13:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-05-07 13:14:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-05-07 12:43:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-21 19:01:23.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-21 19:17:42.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-21 19:35:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-21 19:52:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-21 19:53:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-10-21 20:13:33.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-10-21 21:45:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-21 21:51:32.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-10-21 22:07:33.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-10-28 14:10:36.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-11-07 09:27:55.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-05-01 20:14:49.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-05-01 20:46:56.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-01 20:57:17.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-01 21:00:30.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-07 12:44:12.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"CC"},{"updated":"2026-05-11 16:13:15.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-06-19 19:57:02.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"a423ccd85df9222286f19248630e39c57d72acf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-10-21 19:00:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e098a70fee0a465089afdd27247d3bb7f24e1e9b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-21 19:01:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"856c56dce2327c1f82778501e50c8a2143367a84","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-21 19:17:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I097408e24d8ad05aa10de26551d540c685f9c22b .","accounts_in_message":[],"_revision_number":1},{"id":"aaabe8da1f6483905f7f3e5c4643f98ea500e027","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 19:35:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32eb50f901d7402bafcf24f0f0af8b7a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8192119d737f4e9e867435f2b8725b56 : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/12091215fe19400bb5cd1e1687253552 : SUCCESS in 25m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce32206553ed4a669dd4a76ce92a19b6 : SUCCESS in 33m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d6c8ef43d64f561136e917d5e4907688264a865","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-21 19:52:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/edf8cf4693b7405385fdf70476040b34 : RETRY_LIMIT in 13m 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c3c41c3090954db88702813bd124bd08 : RETRY_LIMIT in 17m 30s","accounts_in_message":[],"_revision_number":1},{"id":"db3779fd03ca1f1e0b1a0f1bde97f091445d0ec8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-21 19:53:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2093/.","accounts_in_message":[],"_revision_number":1},{"id":"36fe32525103c5e41471e3f706aad1ac0a8b1df7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-21 19:55:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2207/.","accounts_in_message":[],"_revision_number":1},{"id":"db42ada503873b599ec7c9f4e77a0acbc1bfba8b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-21 20:13:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/964492 : FAILURE in 1h 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"bce595e35a0c3bdab39a6fd787e046e3e9baef1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 21:45:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83426db2eefb4b68aaa93fc637971258\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79accddaef8647e7ab0bb2a0cc320477 : SUCCESS in 6m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d0b0d7a9bf94c2e82335f8d743d6de4 : SUCCESS in 9m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb413457eaff427fbf70d250a292c2f3 : SUCCESS in 15m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8af956da39a4ea88f6a0313dbda6728 : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/1724b44d87614ec396a7ddef0484a1b6 : SUCCESS in 30m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ae74d23a8e8e468c8e49edf5729a3004 : SUCCESS in 1h 00m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8800288a40ac41a3a84f8c33115dd99d : SUCCESS in 1h 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08dbd6554a09465088b1e504fc854df6 : SUCCESS in 51m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a2cfdcda00dd4065b8173c1f042bafb3 : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e1d4c655c18349edbb3b13ac205b7423 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a75a7b21cb1f4a91911ce74bb7f8458c : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9a79eade73f9444c81ad01194862b445 : SUCCESS in 5m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7feafff73460463db78c32f48756fc22 : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/247dcfae698d432fafa1c57e2b352896 : FAILURE in 21m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8d1a1926be1e4cd88da08ed41393e4eb : SUCCESS in 6m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c1658316dde44217899e83fb2a963d0c : SUCCESS in 1h 23m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4f67a0f803624e10a072ccaf290507ad : FAILURE in 1h 41m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a0a251f09db641c7a409bfb67d4d4d4b : SUCCESS in 1h 40m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/227ef6d1a50d477d87ad2e8ec71e71ed : SUCCESS in 2h 39m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/10387f9b060e4edda971220d476984ec : SUCCESS in 17m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1bc6e8e38bd849b1b0a4f3cab42f641b : SUCCESS in 1h 10m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8849564c675c489d977e77e4303568cc : SUCCESS in 1h 19m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/723beb002d474f1a918d8116729caac9 : SUCCESS in 31m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f12cfa0fd7a84b7786b22baeb5cfe889 : SUCCESS in 2h 14m 00s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/a20e2bee775c4c58887a6049f9e5dba7 : SUCCESS in 1h 16m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79d0e3cccda74ca9bb59e0f8638fd893 : SUCCESS in 1h 44m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f9a1a95bd1f4902ab5106bf6d5d98d2 : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a4113bddd815ab429b86e6d5cc7377f48f67eed9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-21 21:51:32.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_10_2025_21_45_10\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"4d6ad65b08ea914d6789e3000360b4bb55d49d5a","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-10-21 22:07:33.000000000","message":"Patch Set 1:\n\nhitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-92-964492-1-fc: SUCCESS\nhitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-92-964492-1-iscsi: FAILURE\n\tinformation on this CI: https://wiki.openstack.org/wiki/ThirdPartySystems/Hitachi_CI","accounts_in_message":[],"_revision_number":1},{"id":"203fdb1949fa1d86bb6c77a851fd2d4780b85378","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-11-07 09:27:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c480234ca6b0d5a53b1093c3129ada981e3b33a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-11-07 09:34:46.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"614926e3612c3bfdd5375e715e07ea7cf4f17c5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-01 20:14:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c1140fea848f9226e676f94ed952d04e6101878c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-01 20:20:21.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"4057ac19cf74c7716daa6695e7697a40f3a1f7dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 20:41:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32dbdc8c0de74dbd8a951dea5c2a7b4a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e10fe40c89474840903759696a80cc0e : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/42b1c147517940dabd97420af998b496 : SUCCESS in 15m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a9278fba09fc4f9bac7e94c95b64e65b : SUCCESS in 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d3947e2a0b3c1f945fbebec84d8903651722b4c2","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-01 20:46:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver build/74016a6602c44019ae5528a250c22c5a : FAILURE in 31m 42s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/964492/etc/tempest_log.log","accounts_in_message":[],"_revision_number":2},{"id":"6a50e2678de5fd917b1a9d7ad67a316269fb38f3","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-01 20:57:17.000000000","message":"Patch Set 2:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/964492 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"51a864651d621ab88917fcac0ba004efee7573f0","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-01 21:00:30.000000000","message":"Patch Set 2:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/964492 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"3998e2c985a56e5b2e66ddca346324f6d60e908f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-01 21:15:08.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I097408e24d8ad05aa10de26551d540c685f9c22b .","accounts_in_message":[],"_revision_number":2},{"id":"e4ef45377007a292ed999a27c744e2053f6ba835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 02:15:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddfc1458c0cb412993cbfdc55231b171\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8a78aa482f24247bf72fa9e937183a0 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa312c4710364ea8be9abeafa6f28789 : SUCCESS in 8m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f352b84b7f25496792ba03f42002cd8d : SUCCESS in 6m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/afe38b200cc449f58590232e72771c34 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f72597b903a4b1eab0851babc89324e : SUCCESS in 12m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/b7ec6a95d7f4484ea5607ef195155522 : SUCCESS in 54m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d7724b9ba15e49cc8cc421fd49fb2378 : SUCCESS in 1h 00m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/71b420d32af64fd2b4397fd88989312a : SUCCESS in 1h 28m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c094409be8c443bb84167ff0b3af442c : SUCCESS in 41m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2d974ac1d7b4ca6aed57bbc2b5fd3df : SUCCESS in 16m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dd699dcae4184d809e664aad159505fe : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3470936028e54437ae68feb615e4e998 : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f864e04c26194c42838dc2920bb5e2a5 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2aae5c73e4ec40188577da19f305285a : SUCCESS in 6m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f3c6c19579c340b0aa346a5205e6838e : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/26e7c829cc36491eb48b25fe26302227 : SUCCESS in 29m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b21eaf789dac459c9f7eee7fa090f60c : FAILURE in 4m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6c469ce3abd3420a822b0cdca267d24f : SUCCESS in 1h 01m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9adcfc7dd08643f5b4e5cf8e98892d8b : FAILURE in 1h 59m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/899afe962a4d4c8fb6b8133c6aa73814 : SUCCESS in 1h 17m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b2fe437edb1b48d69694c0566b353525 : SUCCESS in 1h 50m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/822750092a824d77b617ea23b266697e : SUCCESS in 36m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/80617983313041eaa0322aca449cb80b : SUCCESS in 1h 14m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/82d288cd9fd34bc28a7b49f8e134c500 : SUCCESS in 1h 41m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0f46a0d3cfdc4021bbe4f5351ec343f6 : SUCCESS in 25m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/82e7ece5c1fa423c8cb49bba8d272186 : SUCCESS in 1h 03m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8bdbf621e4d34c84931c739b39393995 : SUCCESS in 1h 35m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39a147472cf042a59ce7e9bfe6ad8a18 : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b2ccf96d93a0bc58e7ed9d4e7495ecc64c6be46c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-02 02:25:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_05_2026_02_18_46\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"a53ff2e4092b433210b8871fe684fa36ad83140f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-02 08:18:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4058/.","accounts_in_message":[],"_revision_number":2},{"id":"4d0ecfb8fe744bc65e55bb14e5fc23f93410333b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-02 11:33:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3882/.","accounts_in_message":[],"_revision_number":2},{"id":"23d8c5d944a8f38cfd2563b791542dee516a3343","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-05-04 08:37:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d04b3f96e5accd363a02111df2c984b281ffacad","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-04 18:10:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e8bb98f84b6787f22ad5939d1cf60fb1f1da57e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-04 18:20:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fb1ef19d4ad64862490624386119c4cac6a299a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 18:34:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e355dca5e504851853041187c808e45\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67456c098bf8428c82647a1e8f0c7431 : FAILURE in 5m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9b50be0cd8fe42c98d5b33ab58ac4bb7 : FAILURE in 7m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da401cd9386840ddbc6abaf757a5f7b1 : FAILURE in 9m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc738cb3a039a67cfb8086f5d6136f059b902715","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-04 19:00:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver build/ce1762c172e740f18b8ca2e55885dfc6 : FAILURE in 12m 39s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/964492/etc/tempest_log.log","accounts_in_message":[],"_revision_number":3},{"id":"a04bd69dada10d7791f996d9475f6cf6c69175ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 19:11:24.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/841bae66f4a24a5eb7f23591b3d32c8a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c823aadc50d14124bb98d104756787bc : FAILURE in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ccefcf0467c456f915c666e7a30098a : FAILURE in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc2fd9a97f084f218be4e1f187a18b43 : FAILURE in 4m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d2f293fa24a41c88d082024838df552 : FAILURE in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68efaf88a8134cc2b0718441c12eea34 : FAILURE in 5m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/9a2a2c02cd664c37b9bb5339338611b6 : FAILURE in 42m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a57de9eb4a6d426d80c040ae0e113b49 : FAILURE in 32m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/963c074a0add44779c8370864822aa16 : FAILURE in 24m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e4deb33e1224937b1fc89e28d4b62b1 : POST_FAILURE in 26m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e60156914a744aef80b43875b84886bf : SUCCESS in 20m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c3a17f0aceba41428356b3cef49b2584 : FAILURE in 4m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b12ebddf51444262abe5546175cf8b2c : FAILURE in 4m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e7e0552393c6467da5285b5cf667b97a : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a4bb5257631e4f278269f32a2bf0d168 : FAILURE in 3m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3507fb79e87c44f585ba0542ffc5c26c : FAILURE in 3m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a7cdaf4ee28f49f397b6116972767482 : FAILURE in 18m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0389b4d7cf8d4bef9bed8c66096c123d : FAILURE in 6m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1b6b0f8e0e5d4e5bb9de6d7e20572c79 : FAILURE in 28m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2203763f3eb74230a8fbc9036759ce76 : FAILURE in 32m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a3b399af9896457c8a2171b50c7d035e : FAILURE in 27m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1243b16987884d08b537326c4787e7b2 : FAILURE in 39m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/250f7bda2f4247b08704cce4c0842bed : FAILURE in 13m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f74112d740a94b4b93586ba2c1f1c285 : FAILURE in 47m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4649c1c0183247f6b906a527ddf1dbf9 : FAILURE in 17m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5fab0883ef1b4f4da964254136dc792e : FAILURE in 15m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e1deb1f1a6414e809684fc1abe962fbf : FAILURE in 28m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/42dc38cae6fb41ec867699c0f9eb4b7f : FAILURE in 29m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0245495792f64bb48ad4dda1f81bbc83 : FAILURE in 23m 58s","accounts_in_message":[],"_revision_number":3},{"id":"8ab5688d88d29b24b041c2dbb9450772e3749e33","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-04 19:18:57.000000000","message":"Patch Set 3:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/964492 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"cf8c2084bb8133f10aaf9c3f82a4130889b4a834","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-04 19:20:37.000000000","message":"Patch Set 3:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/964492 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c6f3cc042924c9ad8eefff52e60808cae53394e4","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-04 19:30:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ecec441122e73553e31671d1151d21e0609a2be7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-04 19:32:55.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3889/.","accounts_in_message":[],"_revision_number":3},{"id":"e0cce4da0a2d7dcb140dcb1ae542fb2efd9da672","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-04 19:44:07.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4065/.","accounts_in_message":[],"_revision_number":3},{"id":"712d19e77fd98d43d8848121e3bcac9d1e6b622f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-05 20:25:47.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"daeafff1ac3d0a336c14460e5066ed1f3a914dcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 20:41:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/493cc433c28d4106b4df6dca72556d95\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dff6db66f13f4a5f964077409f7f64d5 : SUCCESS in 14m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d6e8a6213fed4b528216f5078b31b3ed : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3b4f16101c374aa4b442e2b0c9d81ee8 : SUCCESS in 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b723cacfc540ce58eb397320d40fb41c89ad578","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-05 20:57:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver build/0abb82268c4a4380a2b650be3c94e87e : FAILURE in 31m 36s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/964492/etc/tempest_log.log","accounts_in_message":[],"_revision_number":4},{"id":"f6bffbcf6d23946f53c8bf27ca874e44fbd10d10","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-05 21:08:22.000000000","message":"Patch Set 4:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/964492 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"ed2eb71ec31823c822bbb1d8cd26c2b5de6ecfbf","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-05 21:22:45.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"46a0e19d5d408c381aabce8012691f66e6dc69f5","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-05 21:23:24.000000000","message":"Patch Set 4:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/964492 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"c8d6b830c515ab857c30abbfdf082de865c3b4a7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-05 21:26:35.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I097408e24d8ad05aa10de26551d540c685f9c22b .","accounts_in_message":[],"_revision_number":4},{"id":"33cf18ef10632ed4d47d385132ebc0eac240502b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-05 22:13:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3900/.","accounts_in_message":[],"_revision_number":4},{"id":"0eaa89e86f5cfb2443b88bfdf3f072319f2b986b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 22:39:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d967e97a52614f219f05771535dd9a7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/628728b8988c4e9db78f62262a9277ba : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4968c42ab6b41119a99af29869dab0e : SUCCESS in 11m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/931575e1b0c14784923ed707d942a6ba : SUCCESS in 7m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c9e68a1a29ee4ac1b77cac7f45c5bf9e : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/967f1235cd1947389142cf0451333e66 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/ed15c2e7ec3d491bacf93fa5a53f7e64 : SUCCESS in 38m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa294b197a7c4c71991c10cc6bb28e98 : FAILURE in 22m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7a54f92311d744bc8dc68c6d7ebdb2d3 : SUCCESS in 1h 25m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe219aff038448f59fa7359583459edf : SUCCESS in 1h 05m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dc5fec16d6a4f9a80bdd85f5c8b1f5f : SUCCESS in 16m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fd6b1166234d418a9752738a2d4e3ffd : SUCCESS in 23m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6570e4bb493844af94430c43091a755c : SUCCESS in 3m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4c66dfd2632d4f408cf20e0594c43168 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b71aace008f74ea48de8a4c233b47581 : SUCCESS in 5m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b273b7c03a2a44e68dc380c5abcf31f2 : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0b7ff1e36c1f47ea8a2257cea0db1aa7 : SUCCESS in 46m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f6957eadc8bd4dca8100c735b9919ae5 : FAILURE in 4m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/761288df46be4532b58cb636647025fb : SUCCESS in 1h 42m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/96b5dab4fd334545964ba9c9fdebe255 : FAILURE in 1h 30m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/adf577a4ec354b53b0ced4956ca507fe : SUCCESS in 2h 02m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6179414b19e74b87b6e8d75de2b59f63 : SUCCESS in 2h 06m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/86a4f8a128a44367baaf8f4f87837418 : SUCCESS in 18m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d436e8fca6274bc1904da79c72550c86 : SUCCESS in 1h 30m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fbcb236bec144871a80147dc9c287621 : SUCCESS in 1h 59m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/64a7e5b4f396473c885b73ef09bc02f2 : SUCCESS in 45m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9afc74fc03a6476082e4f2b7cdf5702f : SUCCESS in 1h 25m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5db26ddfc6b44a23aed29f8dc4f4acde : SUCCESS in 1h 48m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e55f5e04f2a405c8bea199c620386bb : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":4},{"id":"ab53c2eb804e6ed979cc90ae0882d3dcb0e17017","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-05 22:51:14.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4076/.","accounts_in_message":[],"_revision_number":4},{"id":"a1e20f377df89968294f073a2cae0feef0cd296c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-06 12:51:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e51bfb6239c149b73817df00bb5afc4f5bae1f86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 13:13:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd0f21571ff84f54998c7c630bcb598d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9fb3e2428c5e4f49ba52a1b036feab21 : SUCCESS in 16m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/89c567f86fbf404f9ab98be60d17eb01 : SUCCESS in 15m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8e11902e28a647f3accf8045d8e2494a : SUCCESS in 19m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c9873439171213bed7a3413ed5e0f36031c06426","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-06 13:23:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c9feedfffc46b64365d64da16797e39636240907","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-06 15:57:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-92-964492-4-fc : FAILURE in 1h 6m 34s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-92-964492-4-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"61aa78b7114e2e71ef4a86ae05a22dc1b53837ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 16:09:32.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7762dd0ca8f042e78b98d6fed33220dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae6516f7df2b4744bf9ba6fbfb573511 : SUCCESS in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d31e6715a8a94022aa80e837bfc7cd6e : SUCCESS in 12m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11faf3d2cee94e98b3174460ebba75f4 : SUCCESS in 10m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5ac74ed7d420429eb7a92db5c57f07cf : SUCCESS in 15m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da01e668b0654b54a8d76af237ef53ae : SUCCESS in 13m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/691cfdbc8ebf439d8db3940857e8eba7 : SUCCESS in 1h 03m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25425098232a4d8590368ea7b74e016f : SUCCESS in 1h 11m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff5ad66de76c4e7eaa01002973b6a944 : SUCCESS in 1h 18m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecf829a4893d437781d998822043dea7 : SUCCESS in 33m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/948d5169e3814421bf5bd66748624f4d : SUCCESS in 22m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3d155a94244841e4b8cbfd70eac6b08d : TIMED_OUT in 41m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9f1e569921ae412e922dcd32f5aee707 : SUCCESS in 2m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d1af108473574a889a99848ccc3886d9 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1e7db80112f14b2c9a9d736119135571 : SUCCESS in 5m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e37c897f9ddb4a619463aaaacf42eaf4 : SUCCESS in 6m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f6e4b8b86e27402f98fd39c83aefcfac : SUCCESS in 39m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/430c1e7bbc554712a8f99dc35a6884f9 : FAILURE in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aac33e90ce3e4a35a091717fbc0d3b10 : SUCCESS in 1h 44m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ea1b77c9d4b84ff4bef99716b7a6a00e : FAILURE in 2h 35m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a319d036dc544e2589efc1d273184417 : SUCCESS in 1h 37m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3481d8140ff942adadd6b21c6a4c1e90 : SUCCESS in 1h 45m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5f26450956b042aa98f9d8ac6deb6a66 : SUCCESS in 38m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/431de44816374238970ef13abbcac44c : SUCCESS in 1h 28m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bedadfcf2e8c4dccb767a447d8e81906 : SUCCESS in 1h 36m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7dc3cc2ea6264d9fbdd373ecd3cc81d6 : SUCCESS in 41m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dbe2415113d5425c91ac15620d4af28a : TIMED_OUT in 3h 11m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85d8922e468542b29d5ac000a053f545 : SUCCESS in 1h 18m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17ad692fe6fc41b192df295179a3bf38 : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":4},{"id":"0941aa9801fb35fe7157bafdb6bba0e4118a736e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-06 20:28:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"756fff2cd3ebf5d2304abb41f9a3b90d86b1b981","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 21:05:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bd0fd89e2a94bbda30484c632832f80\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/76f358cb6aa4463c9a59dcc97db410f4 : FAILURE in 5m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/26340f776c804de6b1fe7d3f8c1e882f : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f493b04393d54bd0ba8432a93fc90b0f : SUCCESS in 17m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc9c7a4ef5acd32c9d0ddfdf7a1f4db374a481e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 01:57:00.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c134d3d341a9484e879f4e3cae7dda09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3f691387d9e4164aa4bd6816b9bd553 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90d3e31621f44a1593d4875f17fc323e : SUCCESS in 13m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc8532a904324c5da2569a366f3dc790 : SUCCESS in 6m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/319d8c70ca9d4d0286d21481e4f1af5e : SUCCESS in 17m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b4c7266b78431a841be3a70590dbdd : SUCCESS in 6m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/4a41315041744896a6dca1b035cd40d3 : SUCCESS in 1h 03m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4bac38a14a4647fb96803099f7829db1 : SUCCESS in 38m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5e4d8a395fe4970a9716a6a38c304cf : SUCCESS in 1h 31m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4fb4bbd4cb9455da5f5751376e0fe7f : SUCCESS in 1h 14m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b610fd2fee664c7da869d90f183096d6 : SUCCESS in 17m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/32f497437db34e599ce2c4a1d125b174 : SUCCESS in 22m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d94a195cd1bd446d8652c44c5e3c7a86 : SUCCESS in 3m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/194adef3afc542e8b00582a8a48d7dd3 : FAILURE in 22m 39s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/92f9f588e8ba488abedaab8cb179f2a4 : SUCCESS in 6m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7d02376adc30477b8166482475187844 : SUCCESS in 6m 05s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b8c6102e861b4f88b044f0729462f7ae : SUCCESS in 8m 57s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d160136dc8b34d67ba5ca4ff1225e9eb : FAILURE in 29m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3ad769a5df444a778c3cbcd92dea85aa : FAILURE in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f2f5e8d36f843deaba24c39ddaf2d13 : SUCCESS in 1h 43m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f1ce343de7d34f02b57600a7d4140a7e : FAILURE in 1h 37m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/10c4778587194d5fbe43f897395988ff : FAILURE in 41m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/51c38f25a45b4185bcd9f569c191da56 : SUCCESS in 2h 18m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2e41d93b3764487ca98271c94c2a5076 : SUCCESS in 23m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8e1b7eadd0734721bb1b8c69eb6cac9f : SUCCESS in 1h 18m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3054615268584fda9fbff31a456174dd : SUCCESS in 1h 35m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f92b31f18a1f414f973400da48821ecc : SUCCESS in 44m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1ef585d92cd949b3953aa573df4576e8 : TIMED_OUT in 3h 09m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a08b62fdcc0a4975a1557ad4ffa85e14 : SUCCESS in 1h 40m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cedf36679a8c473f824b1a3a74c0b995 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":4},{"id":"2e8b6e9494f0781e133fd2800149dbd51be9cf22","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-07 02:36:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"47958b61b2d0e32b4a5c1049ae3ed4968fa1a4e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 04:35:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e916bb3dbb94deebfc345f9e8f96a9c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa35c4b70bd04ddcaed875f32f5b5d93 : FAILURE in 7m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3f104674390443e69c191ef59db9940e : SUCCESS in 14m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9af926b4c2e14bd58ffd9ceaadd25564 : SUCCESS in 12m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"189b1bfdeaa2eb1cbdfd321922d7fa434865c0b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 06:31:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f45ed6143eba41ecb3aa21405302c064\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3793c3e584b1423c861d8508cfe25bcd : SUCCESS in 7m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87c6d9563a4f4e33a3177dcc4c24fccd : SUCCESS in 11m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/135970ce60db4045be0cc0b8aa749aa5 : SUCCESS in 10m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fc885d6f2bd94e86bdb42d6b8d7fd080 : SUCCESS in 14m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b2f63cfe914d588c661a63af41fa4d : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/05d84a108b8f44688908935a9e316943 : SUCCESS in 1h 13m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14214adcfe9f43e080db20d9e9b0b8ff : SUCCESS in 1h 12m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc4130710de647808a7feb22b0d4a715 : SUCCESS in 1h 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01d7caa2963f4e4287bf02d774b72fa4 : SUCCESS in 1h 06m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc193a8844b84b33bb311ebfca8a0dd8 : SUCCESS in 31m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bdfcdd359e8640228afa28fdd6abf8c7 : SUCCESS in 25m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1c57668c69614eaa85f0463a8b25c973 : SUCCESS in 3m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf4d72abc84e46538d4163edc58a8909 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/826f85f9c8484ed691a9557126ab0fec : SUCCESS in 6m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c369caf46b874f96b44fd874bee373cb : SUCCESS in 6m 14s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/aded7a3cb37341d1bfc3f3cc4f411fc6 : SUCCESS in 9m 30s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/474d76d098714cfea93c9a46e689faa1 : SUCCESS in 36m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/42e10dcdf8b54a40990389835afe6784 : FAILURE in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b1ce6d77bf864709bc13083310e79f44 : SUCCESS in 59m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1ba0154ec1134502b8c8f8d77904a0b7 : TIMED_OUT in 3h 28m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6c140bb3633a409aaa79eab4738f206f : SUCCESS in 1h 49m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0240c96d68eb46109a997de90d007390 : SUCCESS in 1h 41m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5eb7e272c7a449e7b55dc4886458c1af : SUCCESS in 22m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/523bf383478a4372893107b0b96ff7d6 : SUCCESS in 1h 03m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fde7171d36304bd89a4ecaa3c96ba388 : SUCCESS in 1h 31m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38a5047596734b83901dc8f49e1f6856 : SUCCESS in 44m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4ec23417deab4c6992599de895212ba7 : SUCCESS in 1h 58m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b27c09b752e24d37b7dd33753d2a32e1 : SUCCESS in 1h 44m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c3abab7b4fe49e6bab96f6db4574faf : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":4},{"id":"2ea7356b4874a27605eed0dd90052a1e878c3244","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-07 06:40:42.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_05_2026_06_34_39\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"485a05a96857bc7e3eadbf449c6fbe7ad83553d7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-07 12:43:08.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"333573b53634a837373e9d826f3cca564500aa49","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-07 12:44:12.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bdc19f224934ab63a1a38541a20dc25b600cf32e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-07 13:14:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"927cde53c9e12b341e7b48a94130b1c8cc7b3631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 13:16:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"aabb0a681eb381f4afe42a4c3ec2a1ab32deebd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 15:36:07.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30234fb3c75b4bbbadb0ba9b48cbed4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d75e6a9eeb845dca3b671b1a501a514 : SUCCESS in 7m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13b9b8ddb5b5477dade05bdceeb83ec1 : SUCCESS in 11m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a65dee28f11417faf51e20a68e2e099 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31621830ecbf453f87356fb18bcc36fe : SUCCESS in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/9c8a3cf27c834b92a648b68a3bc5f656 : SUCCESS in 56m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/026918915ef748709429f24428ea7c44 : SUCCESS in 1h 03m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e3f0e237dc524f69b144be0cf017285e : TIMED_OUT in 2h 07m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4339b020bbe4476f89d2928576be71b9 : SUCCESS in 46m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a63e754ca434d6bbe48b001bbeb1a5c : SUCCESS in 22m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/50014c0da9df4c62a97e32cfa8d67ac4 : SUCCESS in 1h 16m 11s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/132fbd4810a9446bab8fe5f7503a2781 : SUCCESS in 1h 47m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d05f707b3ef94f388194db379b815e1f : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":4},{"id":"7c8dbc64fba66dfd4e64324ccf0ba02f06d41d3d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-07 16:28:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"242ee1dd6b7a28a12e47b2cec33c1120f9247acf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 16:47:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cf1e926f5eb464da146b02e67621c20\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/231de008f0ae4498b578998dda129ea3 : FAILURE in 6m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/405a071e542d45f792402d3fc3ff62af : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ac6ad14c2ee841b5aa45adc7054cd88c : SUCCESS in 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7d98395bcc592aa72fb6493bfdba82156d4106b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 03:44:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68059d5bd9104a8a9cf4d211144de4e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c25acdfd78c4e1db351f0ecdb7fb3a3 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08d4c52e98b54701815ab0ce1b6b2989 : SUCCESS in 11m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f22df0d83eab49558e02a6ddf62514d5 : SUCCESS in 11m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/baafc414a3194009862036546a82cf83 : TIMED_OUT in 41m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37fba3ff1f174993b55500247dfb288d : SUCCESS in 13m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/e1196c5e5054489ab93b5169575a23aa : SUCCESS in 1h 00m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ce84c49430244f0b148d84d0a39bed1 : SUCCESS in 1h 07m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7dbe58c0eeb54bf685047d992bb27f73 : SUCCESS in 1h 53m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e488a45b402452dbe3920f03dcf7952 : SUCCESS in 1h 01m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3083634ffd4545e4bb73c1f02ff8d838 : SUCCESS in 20m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4360e1ff1dfa4710a859be7a0f66a348 : SUCCESS in 16m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a90172e3cba74429b63ae49742a74842 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d3102d329ae24843b03227aff3eb5179 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f4e9f226a382417fae2e36f3c2e2606d : SUCCESS in 6m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/876a95ad9e354929bcd6f43b4339de9f : SUCCESS in 6m 37s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/632dffc91cea4874ac4dbd200d6c8c3d : SUCCESS in 10m 23s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6b7c41e12bd04d6e9bcfb4ad82f881a3 : SUCCESS in 41m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e6a449c571054b06af6950a9e0b2ab1e : FAILURE in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f585cfb2e9dd418ea8d7c69b26b8159e : SUCCESS in 1h 41m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b50b9ca3d37840278c156c8af38821a9 : RETRY_LIMIT Host unreachable in 3h 33m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1800fa1b1e0441dd862b4ff13ea1dc6c : SUCCESS in 2h 10m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/155b381dea2e4cc8934af29ae60a2290 : SUCCESS in 2h 31m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f070c232627b4e8b9928f878afe8d231 : SUCCESS in 29m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b10124668ef844f4b46b611ef6b22933 : SUCCESS in 1h 15m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d73a071f7fc34890841f1be7ca98d285 : SUCCESS in 2h 07m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c9ed34c9b05e424494f02a2c21b3e7c7 : SUCCESS in 44m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bc9c50184ea24aaaa14891ef9f03b847 : SUCCESS in 1h 54m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9da3053aa0d24dfa81edd91399600ba3 : FAILURE in 2h 14m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd09fe844800472cafb3504e0ade4abd : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":4},{"id":"e6664249c83891c9e84add1ae288e6edb661a434","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-08 16:35:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e356d1885245b2c050bbe75e866b1345af3af66b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 17:52:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c50a398b4634bcaa0c3a6b2e804d89c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c16af1941a1b4b5db9fe17e67d4c0cec : FAILURE in 5m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f1b3bbb28b0a4d59b4a41edac2651923 : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a6cb86e45d7b45a08e801b5138e39379 : TIMED_OUT in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0f8ce92b918c5870a559d89463493e40656a3cc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 00:05:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/476cbb41556b40f582670acf845195da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b3e2c394a59471390d22df9816de86b : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c3ca1036e2843b1b2f76aec5a680971 : SUCCESS in 9m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbf74d44d899416fa9c1a9ec67a89ffd : SUCCESS in 8m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cbf7c51032a84e27860071c561fc9980 : TIMED_OUT in 40m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ceaf0647554e26928f4b789a354b3a : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/a5627d3484b24d40babf3a015896e67f : SUCCESS in 1h 01m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e034a33fca6d4d29a8c1b649a056a3cd : SUCCESS in 1h 11m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f2f42c8d86e47988a52b39c00cc4cc4 : SUCCESS in 1h 15m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/558ab58e72d24c2bb759c4a17cac139d : SUCCESS in 51m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11b6e1d972d041bdb228063863e03edb : SUCCESS in 18m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/39eecc06ad194305b897c36b1d7b82ab : SUCCESS in 19m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fa7dad62b0d643eea4ba295d7b2129a3 : SUCCESS in 7m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5d627e4db19f41d8ad518badc4071a7b : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4b02f4af0e2b462196ec5fbf0e01d3c3 : SUCCESS in 9m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ec28f5cb7c69474a88126f25fbda7cd5 : SUCCESS in 7m 19s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0b3386040cf64485910d78244fbc0ecf : SUCCESS in 12m 28s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cf8a29a753584a2784a5ad7c4e59a9cb : SUCCESS in 19m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93cd9dad03f943c78fcc1f60762547bb : FAILURE in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/67c3473c68b74307b880d9d967a22f8d : SUCCESS in 2h 03m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ffe3c35279a9481a8337570f146d68ea : TIMED_OUT in 3h 38m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d8325789d4e94b2b93dae48c2bcb6ff5 : SUCCESS in 1h 54m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/be3c8755302e4f38b98e7cf50ef312c5 : SUCCESS in 1h 50m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8af302e52eee4c3db3f3236c44880c18 : SUCCESS in 43m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4826471f6c9b4f30b06540de002b1e36 : SUCCESS in 1h 14m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1b975b3df33347a9afbebbd50315d898 : SUCCESS in 1h 27m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/86bd210872ee495387d7729cbba7f2fe : SUCCESS in 45m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1ad23a8db5214e0e937890161347a196 : SUCCESS in 2h 34m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/84067fcdb94a4e0d811641243f0dec54 : SUCCESS in 1h 40m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36a02a56a97843668cbda4b66a660c21 : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":4},{"id":"bac107fd3f1878d134541ccc1ae24b00f5c51ce2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 00:06:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1ef5d6cd6c293657ac01e01cac1853c4ab6275c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-09 00:15:29.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_05_2026_00_09_15\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"14d9717253bde210292a5d3140d5917bc732d391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 02:06:33.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d71313bedd20408fbbd24a563181297a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98fd91c6d251434ea2ee379a86604009 : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/097b18a3edda4a60a04354adbd9fec44 : SUCCESS in 10m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5edc88f933b34012ae941ef461f4fa3e : FAILURE in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ac4c0237e1545ccbd655850dbc0bd48 : SUCCESS in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/44bb343f9d784b15a7299ec37f32d6f9 : SUCCESS in 1h 07m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b25139540d0e48faac08f3e3abe1d00f : SUCCESS in 52m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ffc22b8613e45f08c0f06b56b54d5a3 : SUCCESS in 1h 38m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f6df40cb5d744da9d5a9928231f2659 : SUCCESS in 50m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bc6a95972a14447acd57fbfc6a8395f : SUCCESS in 10m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4300d3e36cc64d1ba1cad3ac2ce37c9a : SUCCESS in 1h 07m 32s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/85a6b4cff52544a79a5ce785a3eed1a1 : SUCCESS in 1h 54m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9808abc590d44b8bb427a6581ff58fb : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":4},{"id":"f3e77c50cd504753fd1559f161bc232a94722410","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-11 13:07:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5b2545d4e592bebec31c0f197878d3f55f2cbb24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 14:10:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebb29e841ad9401aa11b16ca67b9c83f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb44128126cb4a9e8c7aa7a646119467 : FAILURE in 5m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/066440dd56454cf7a89cec9a6d0f16bb : SUCCESS in 14m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c3cca075c450490195c6d93e521d41ff : TIMED_OUT in 1h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"49c5dcb701266b28073c1b63562221dce6543295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 15:21:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f35bb526d0647bcb4b0690a1baca20a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c70035690554e8ea3a541a29b694eaf : SUCCESS in 7m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf64eac4bc1948edaae3fba2da6d1b37 : SUCCESS in 11m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f576eb0334b548eeb6c2cabab910f368 : SUCCESS in 11m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3894b4f4c86a4231b1ef764eb1f5242f : TIMED_OUT in 41m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0661deb103d44eb08dc236f52f9f3f2e : SUCCESS in 14m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/0783b90948f6486c95a8a332462decb3 : SUCCESS in 51m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/37f02c7dc2a54257b28f2feb6a1d8948 : SUCCESS in 1h 14m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90127f42a90842e29d66b7cfce91427c : SUCCESS in 1h 32m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f071555e39d94ac2a993c6ed7447e35d : SUCCESS in 1h 08m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae4123b4779f40d98fbd967ff6b8b6a8 : SUCCESS in 11m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cce5b696b2274145b317902425b0fdd0 : SUCCESS in 21m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8099bdb10d9840c7b1d4224d333e2da5 : SUCCESS in 3m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f04b483459f448b093b64402f42dcf12 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/132bbbea3d4349228dad175b69cbff08 : SUCCESS in 6m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5e76f63091334e0699def0bc4d5da788 : SUCCESS in 4m 40s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/97ae133ae7ed4531b5e0f1038e4a8537 : SUCCESS in 9m 39s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b2f3a782e5ca438984ce5d37adf560e8 : SUCCESS in 40m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f35d478b0d14f4bb4f886c65f19ef8e : FAILURE in 3m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/219b56ee9acf4d9b9f43f2625538cb5c : SUCCESS in 1h 32m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b07cae45ddf148f8bbafb011edbfb34e : FAILURE in 1h 37m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5422aa85baa24051831678132063253e : SUCCESS in 2h 07m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bfd2cb55c16b44f68e7c48e99a71db67 : SUCCESS in 1h 56m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/75a8b78b43af4ee596c256e27ecffbce : SUCCESS in 48m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/618cc100e09241bdbf3e0c4f511572df : SUCCESS in 1h 24m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6bdbc5f3615452f9c48919d27a14f52 : SUCCESS in 1h 37m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/172d2668a22749f5a9700e1015d0640f : SUCCESS in 56m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5247534add634391a917add2d32b6b35 : SUCCESS in 1h 49m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5b5cf4ff8c94c74b7865e07f8dad81d : SUCCESS in 1h 40m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bde9bddc50de4d4c985ea40daa24dbaf : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":4},{"id":"6993cb00cf41a020f04ea1fd4fbf8ba59d1a261f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 15:23:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"82c9acaf9caa07c1abe03a51f5223ec0d0e6deca","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-11 15:49:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"976af9c8e7702dcbb8b1f3ab943c5d2469abee93","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-05-11 16:13:15.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"be2d4c531fa9f36a335bcb5a5f46b7f576c5dd59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 17:48:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83d13ec41fbf45fe85abdc1c1bd7ee07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0edee6e74944758843ef1aa60dace7 : SUCCESS in 7m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/951e794f2ff343919332af9109b97544 : SUCCESS in 12m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac418ba879f543fea041593232ac18db : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c0fed0387e415ab198acb73d917802 : SUCCESS in 15m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/db5e81e906bb4fc0b51df70f4ff2d5a1 : SUCCESS in 1h 05m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eae514931bbb477b857c7ef4897eb470 : SUCCESS in 54m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/46c4b920a0094e45afcb8518efc53a0d : SUCCESS in 1h 30m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9b544b720474b108cd928bb63ff264b : SUCCESS in 1h 07m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c10be1198fc4861be2be2c35c83e479 : SUCCESS in 15m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2d0752fd5254446685d9ea10e6e8054d : SUCCESS in 43m 30s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a3eca9668ee14e64a51facd2f8e73592 : SUCCESS in 1h 50m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f6f8d12c6d54ca59fcf2c83afba9696 : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":4},{"id":"eb2f2b29e177bd0619628431d1e58596fffcf634","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 17:48:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e4248b66adbca64b9739752ffd7010b475ea6ecd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 17:49:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4c7ae9eb5f24df288cedb1a3bc54f90\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5534491ce143410a84cbfb235aa06b2e : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2e2525a68ad48e99e983d0c05a2e267 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":4},{"id":"ea8dbffc9ae2d86f4ad465685160ebb6afb5da7f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-15 23:30:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/98f79dc3e60e4c6db99eb4ce0c5539f7 : FAILURE in 28m 51s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/964492/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/964492/etc/tempest_log.log","accounts_in_message":[],"_revision_number":4},{"id":"166a0330c75f2c9922dd44ddc5b97ff4f6a9f167","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:57:02.000000000","message":"Patch Set 2:\n\nBuild Failed","accounts_in_message":[],"_revision_number":2},{"id":"0c8c9f23924b11e6d09fc5fdfa85d90b21a3a936","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:58:16.000000000","message":"Patch Set 3:\n\nBuild Failed","accounts_in_message":[],"_revision_number":3},{"id":"93a0031be8695c824b85726925b4855143c3ae0a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:59:55.000000000","message":"Patch Set 4:\n\nBuild Failed","accounts_in_message":[],"_revision_number":4},{"id":"012a396dc9df450354c9e6e19ce6bfc20e9186b2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:03:50.000000000","message":"Patch Set 4:\n\nBuild Failed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8c841d04870c1815ef33ec7fd57bb3f22ec5f5ad","revisions":{"c1292e46543ac7f0c8b5acb6af879ebb7719f8d6":{"kind":"REWORK","_number":1,"created":"2025-10-21 19:00:30.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/92/964492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/964492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/964492/1"}}},"commit":{"parents":[{"commit":"d02171164bdd702b12b59888b744d172f30d712d","subject":"Merge \"docs: starting iscsid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d02171164bdd702b12b59888b744d172f30d712d"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-10-21 18:45:34.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-10-21 18:45:34.000000000","tz":-420},"subject":"Properly set shared_targets for all NVMe-oF protocols","message":"Properly set shared_targets for all NVMe-oF protocols\n\nInclude transport specific NVMe-oF protocols (e.g. NVMEOF_TCP)\nwhen considering whether a driver\u0027s storage_protocol is an NVMe-oF\nprotocol.\n\nCloses-Bug: #2129271\nSigned-off-by: Alan Bishop \u003cabishop@redhat.com\u003e\nChange-Id: I097408e24d8ad05aa10de26551d540c685f9c22b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1292e46543ac7f0c8b5acb6af879ebb7719f8d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1292e46543ac7f0c8b5acb6af879ebb7719f8d6"}]},"branch":"refs/heads/master"},"2902e3bc5bd64595c61b691ad224d38e7951bc66":{"kind":"REWORK","_number":2,"created":"2026-05-01 20:14:49.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/92/964492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/964492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/964492/2"}}},"commit":{"parents":[{"commit":"65bba6830d578d7fe3138dfa65228dfda19eb49e","subject":"Merge \"Add support for AutomatedFailOver policy with consistent replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65bba6830d578d7fe3138dfa65228dfda19eb49e"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-10-21 18:45:34.000000000","tz":-420},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-05-01 20:12:25.000000000","tz":-240},"subject":"Properly set shared_targets for all NVMe-oF protocols","message":"Properly set shared_targets for all NVMe-oF protocols\n\nInclude transport specific NVMe-oF protocols (e.g. NVMEOF_TCP)\nwhen considering whether a driver\u0027s storage_protocol is an NVMe-oF\nprotocol.\n\nCloses-Bug: #2129271\nSigned-off-by: Alan Bishop \u003cabishop@redhat.com\u003e\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nChange-Id: I097408e24d8ad05aa10de26551d540c685f9c22b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2902e3bc5bd64595c61b691ad224d38e7951bc66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2902e3bc5bd64595c61b691ad224d38e7951bc66"}]},"branch":"refs/heads/master"},"2c5c26f08b7f8cee5695c24b9f3228ebf190129f":{"kind":"REWORK","_number":3,"created":"2026-05-04 18:20:58.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/92/964492/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/964492/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/964492/3"}}},"commit":{"parents":[{"commit":"65bba6830d578d7fe3138dfa65228dfda19eb49e","subject":"Merge \"Add support for AutomatedFailOver policy with consistent replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65bba6830d578d7fe3138dfa65228dfda19eb49e"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-10-21 18:45:34.000000000","tz":-420},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-05-04 18:20:08.000000000","tz":-240},"subject":"Properly set shared_targets for all NVMe-oF protocols","message":"Properly set shared_targets for all NVMe-oF protocols\n\nInclude transport specific NVMe-oF protocols (e.g. NVMEOF_TCP)\nwhen considering whether a driver\u0027s storage_protocol is an NVMe-oF\nprotocol.\n\nCloses-Bug: #2129271\nSigned-off-by: Alan Bishop \u003cabishop@redhat.com\u003e\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nChange-Id: I097408e24d8ad05aa10de26551d540c685f9c22b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c5c26f08b7f8cee5695c24b9f3228ebf190129f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c5c26f08b7f8cee5695c24b9f3228ebf190129f"}]},"branch":"refs/heads/master"},"8c841d04870c1815ef33ec7fd57bb3f22ec5f5ad":{"kind":"REWORK","_number":4,"created":"2026-05-05 20:25:47.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/92/964492/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/964492/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/964492/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/964492/4"}}},"commit":{"parents":[{"commit":"65bba6830d578d7fe3138dfa65228dfda19eb49e","subject":"Merge \"Add support for AutomatedFailOver policy with consistent replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65bba6830d578d7fe3138dfa65228dfda19eb49e"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2025-10-21 18:45:34.000000000","tz":-420},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-05-05 20:25:44.000000000","tz":0},"subject":"Properly set shared_targets for all NVMe-oF protocols","message":"Properly set shared_targets for all NVMe-oF protocols\n\nInclude transport specific NVMe-oF protocols (e.g. NVMEOF_TCP)\nwhen considering whether a driver\u0027s storage_protocol is an NVMe-oF\nprotocol.\n\nCloses-Bug: #2129271\nSigned-off-by: Alan Bishop \u003cabishop@redhat.com\u003e\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nChange-Id: I097408e24d8ad05aa10de26551d540c685f9c22b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c841d04870c1815ef33ec7fd57bb3f22ec5f5ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c841d04870c1815ef33ec7fd57bb3f22ec5f5ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
