)]}'
{"id":"openstack%2Fcinder~964558","triplet_id":"openstack%2Fcinder~stable%2F2024.2~I9db121ac8ea631ae2043461929af0ee02ca05e85","project":"openstack/cinder","branch":"stable/2024.2","topic":"I9db121ac8ea631ae2043461929af0ee02ca05e85","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2025-12-12 16:43:51.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I9db121ac8ea631ae2043461929af0ee02ca05e85","subject":"HPE 3par - skip license check for new wsapi","status":"MERGED","created":"2025-10-22 16:12:24.000000000","updated":"2025-12-13 08:34:24.000000000","submitted":"2025-12-12 18:08:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964558-I9db121ac8ea631ae2043461929af0ee02ca05e85","cherry_pick_of_change":961342,"cherry_pick_of_patch_set":1,"meta_rev_id":"f98dcb4a436745f05b5ab792dcb0361ab1875fc2","_number":964558,"virtual_id_number":964558,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-12 18:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-12-09 14:02:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-12-09 07:17:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":2,"date":"2025-12-12 16:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-05 14:20:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":1,"date":"2025-12-12 16:37:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"}],"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},{"_account_id":38317,"name":"Dhirender Singh","display_name":"dhirender-singh","email":"dhirender.singh1@ibm.com","username":"dhirender-singh"},{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-22 16:13:35.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-22 16:30:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-22 18:39:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-22 19:21:05.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-10-22 19:56:42.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-10 11:17:05.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"CC"},{"updated":"2025-11-10 14:23:33.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-11-12 17:21:23.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-11-13 11:27:18.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REMOVED"},{"updated":"2025-12-04 05:25:13.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2025-12-05 10:17:08.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"state":"CC"},{"updated":"2025-12-05 10:17:08.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"state":"CC"},{"updated":"2025-12-05 10:17:08.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-12-05 10:17:08.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"state":"CC"},{"updated":"2025-12-05 10:30:38.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":38317,"name":"Dhirender Singh","display_name":"dhirender-singh","email":"dhirender.singh1@ibm.com","username":"dhirender-singh"},"state":"CC"},{"updated":"2025-12-05 14:19:13.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"state":"CC"},{"updated":"2025-12-05 14:20:24.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"state":"REVIEWER"},{"updated":"2025-12-09 07:10:16.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-12-09 07:17:30.000000000","updated_by":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"reviewer":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"state":"REVIEWER"},{"updated":"2025-12-09 14:02:19.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-12-12 16:37:05.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"12e4ce4162ea2b4f9d49aa2f708aa9970a407a30","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-10-22 16:12:24.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.\n\nThe following files contain Git conflicts:\n* cinder/tests/unit/volume/drivers/hpe/test_hpe3par.py\n* cinder/volume/drivers/hpe/hpe_3par_common.py","accounts_in_message":[],"_revision_number":1},{"id":"35389447bd6597b4eaccac72d3ddfad7ff589d89","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-22 16:13:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"6d9a4072920fd0e2d28f840e82f162afec200451","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 16:30:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b59973f6a0a24d139c883848c8179648\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1330c123851b48a8b1b1bf27e11e7169 : FAILURE in 8m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c0ab0cd4ef1048dab0d7cc4828e9ff6e : FAILURE in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4805da0dee80bddef65be2d38744aad6e3f98a1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 18:39:11.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/869644da3e8b4a528619aeb2f9f1b811\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e77834d9544347a345e29acdcdbb2f : FAILURE in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eaf8012c18674ee0a97ff180bb1ed62c : FAILURE in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7893a19281cc49df8832f82c723b0769 : FAILURE in 5m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fee8e404521348a8a0fcc000ab9ba9eb : FAILURE in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f75c06d19e7d448da135b46cdccfc064 : FAILURE in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/b3520e31f93e4115bfdce99dadfd9bd3 : SUCCESS in 57m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bfb31cfe7109400dbe50ee4bddfb4100 : SUCCESS in 1h 22m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fde9ce309f14e12bac844773ffcae86 : SUCCESS in 58m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/088e6fa991064172b8c73bacc2254adf : SUCCESS in 12m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1a6caf2d6f774e44a7e1a779060418fa : FAILURE in 4m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ed64587c2aad4a86bee850a2a59f5bf2 : SUCCESS in 6m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/df1c4fa4074e4fceb51c654d4b40f556 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/59fb900d87e448308ec35d5aba789304 : SUCCESS in 5m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3541ddfc85394a5bae8aaafab71a5d92 : SUCCESS in 6m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/409a515f826f46ebb2eacd4d731e719c : FAILURE in 32m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d9682ea8bd8457f9853cb666320969e : FAILURE in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/080b173573d84d3c8ae82670d2b2d6c2 : SUCCESS in 1h 00m 26s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d027c8e6dd2049b5b4edc9d832c826b7 : SUCCESS in 1h 21m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d258e59152a2478990310c8d3fd5c9dc : SUCCESS in 1h 48m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/efefc8d710554ca6b2cf45100c9bd2db : SUCCESS in 31m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8ccbe74feac84ffcac994af1eb1dbbd7 : SUCCESS in 1h 09m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b3c49e7bbae34649b5a8b91658b0cd46 : SUCCESS in 1h 15m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/14c99483e0854467be3aa278dc6b1f98 : SUCCESS in 40m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d307b83fe19341ddbe874a70036dfc9e : SUCCESS in 1h 43m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c3fefaee24e843a6aae00e4626145986 : SUCCESS in 2h 01m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bf64716fd7d4982b93182996f80fafa : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bebd705b9eaf1189d398cd650860ed1b8a0e6e70","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-22 19:21:05.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2101/.","accounts_in_message":[],"_revision_number":1},{"id":"051d518707aade0510dab8d2d43e13256e37f222","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-22 19:56:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/964558 : FAILURE in 34m 58s","accounts_in_message":[],"_revision_number":1},{"id":"841dae66d149ab9583dfff5418108769c8cd7807","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-23 00:31:34.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2215/.","accounts_in_message":[],"_revision_number":1},{"id":"373b1eeab6f09f2b9c01b4b148df96d2d34e7b33","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-11-10 09:49:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e8c0e414aec597cbd800f2831e49713376287968","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-10 09:50:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"d0534689415470cf133182dc15d0689be589ecc8","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-11-10 09:51:21.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"61694903f7fc47b2d20dc3632993c89f30258f37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 10:09:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfbdc8456b994e0ba7fd88b43671ace3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74cca36b579c4f2f9da1ef3bbfd12da6 : SUCCESS in 14m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eedd8d7d17a645a5b4739105168b0130 : SUCCESS in 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"467817d16c1d88153acbd86952af8516b7130cd7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-10 10:18:37.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2351/.","accounts_in_message":[],"_revision_number":2},{"id":"57e2af4adbe446ad41525e694bd284a2f8716d8b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-10 10:24:19.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2465/.","accounts_in_message":[],"_revision_number":2},{"id":"5923b4c61248c3f21171732d1e38526430750c68","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-10 10:45:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/964558 : FAILURE in 37m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d89b8611b4b4e611a53fa5c6d8f2e86b264cf521","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-10 11:17:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd0f707d35b7d5b5cc45d80ae4963fa7583d386a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 11:45:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3b72a62ba174c85b3a828d1b95c0c9d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb43015b9876464d9ecab8f2fe1ba387 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1707536c0a424ba4b34311826cbb5627 : SUCCESS in 11m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b79bce1e802a4f6c8110117d16a56eaa : SUCCESS in 11m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/814f385e1b4d4802a7efb5932a779c41 : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da3cf6e910b1406f97b826154c485cb5 : SUCCESS in 13m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/4b4d14d1118242839c66563cda378160 : SUCCESS in 58m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44306e64ac0a4fa390abb72e81c50ce7 : SUCCESS in 1h 15m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dff13f479dd549b6a110bc169c1ecf12 : SUCCESS in 1h 01m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e236b6d3c48c400188e4306ac4ec0766 : SUCCESS in 10m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9429d8e87d0246e49f4309e9b8336a7c : SUCCESS in 12m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d5e975f2a216468599af06654ad00ad1 : SUCCESS in 6m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/32ec2835e24c4eb98299da0875f6b6b6 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c03c85c6ff249cb8d776fadb0fbd99f : SUCCESS in 6m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/47a50b5c9cfd489192e78823d42aff32 : SUCCESS in 5m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e57c0c33fe1343a3908b29510c21a815 : FAILURE in 23m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c48d032fa994f209600ab77f2f9caed : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/472d4ea2cb504c3fa5418141df207693 : SUCCESS in 1h 38m 35s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/57c9ee544268435bad8810997287ad76 : SUCCESS in 1h 41m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d99f5eb5fe5e415ca30e2dc09c92b5cd : FAILURE in 48m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cbb60a7a2ad6445e8b5ae26002539bfe : SUCCESS in 31m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a4c2dddb18ac4cfb9999c769f6ff39ab : FAILURE in 45m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ca598e1495714c19a66a305787ca4898 : SUCCESS in 1h 24m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6a06753135704215a90d13a8fc33ff4c : SUCCESS in 38m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df0c7b9097da422a9197970e176b6f3a : SUCCESS in 1h 47m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c50d28b8167c4cb5876e6616f31b22e0 : FAILURE in 22m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d295495ef2c4350bbd558258285f77e : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":2},{"id":"aad66303f51253ce9f4ea90f72c8b859d9a8a7bc","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-11-10 12:44:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da6b489f53216c5e3f542f028d779c8d2efec564","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 13:05:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96dc88ad88584eedb47e2505fd024a7a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22bfa5ecca7a45f5babb7ca9e087471c : SUCCESS in 16m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c1f87949d3d449d297672bbf9dc0b70f : SUCCESS in 18m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54f2a0c1a169a5498aa5bd114064eaff33273ac2","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-10 14:23:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-964558-2-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-964558-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"aae9b1721a1f7ffe46d28672a1344eec9902fb46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 14:56:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0d611ce2ebd4ee39deb7de11813d179\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af81aac667b54e1590f91b04997e36a2 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d23d982f8ca842abbd331fba65e0b0ea : SUCCESS in 9m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe215ae3605140d8a1724e143d927146 : SUCCESS in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b56b1d716f654a17a7fbec0862008cfc : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88f7971818b940f48cf07ef2753f09ab : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/7c63b2e02ca14ef78d2a053fdcdaf0b9 : SUCCESS in 57m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/65333e0abbe64e45bce8d7d0a6977aad : SUCCESS in 1h 23m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9684d75d9fa44a459b5a21b1e8c4e355 : SUCCESS in 55m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04950e6e19f7403b8e47124fca2e57ee : SUCCESS in 16m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d58ce4f56d34458aab1e4b73d498752a : SUCCESS in 16m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/930ca693486c44bda8aa6ce84cb6a069 : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f1c9df57eee043f5aeda755767f9e88c : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af72a099ea9248798050297fdaaf7388 : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bfedfc2637d64d0f8fed391020dca5bf : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4a0df70d1bd84206889ac56ace6b9502 : FAILURE in 33m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fb30641f82214212bc9d8d784c522457 : SUCCESS in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d94d0318f26048bdb975bb2594cb5145 : SUCCESS in 1h 47m 51s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0abe994096be4d7da6efb3bee45f0236 : SUCCESS in 2h 02m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5fab88dc9e404fd3a2c09ec889185b38 : FAILURE in 27m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b801b28ec51c4831865c5c5ed5dc2ae6 : SUCCESS in 32m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3ab2c72683854c9381f92450b1861f5a : FAILURE in 43m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b2b95467650347e5b47429a66e453ecf : FAILURE in 1h 47m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/561d6291fa0640bdb9ccbc149e4199fe : SUCCESS in 41m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7ed9cd4b1fbb4227b2953f87e7562f2e : SUCCESS in 1h 49m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dbd4fbe3f9db4485afae635f36171ac4 : FAILURE in 23m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fd7d908aab743a09b8b102c3e2f4fb5 : SUCCESS in 51m 44s","accounts_in_message":[],"_revision_number":2},{"id":"49f10266d832cac08f02e8ae2a50a841b9fc940a","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-11-10 15:32:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a133c0fefd4add8480fbff9275134d2da3b93e0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 15:51:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29ad596f8538418b951b86e58fde704e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f2744606e194cf8843180452b379c4b : SUCCESS in 17m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af36cecdf4344c12a8c11ac4fb7d07d2 : SUCCESS in 12m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d8a53e8061591e883b6dd1375b9f5e7393ed5e9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 17:41:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99aa81594a6b4a29aa49329b6d3ecb6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2a8ea21a8f418c84241ee576e7f415 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b62d9eaacdd4eb0b3a63e4a135166ff : SUCCESS in 11m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f18a56dcde64934a5b2a8536070f509 : SUCCESS in 12m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f5aa133b3104811a0c50ffe8681fc76 : SUCCESS in 14m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9623ba3454f54f9bba0347403d0eec7c : SUCCESS in 12m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/1f1b74868a8c45478bb5cc0ed3102421 : SUCCESS in 47m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e05118b33f04b629624cb9b01d3478d : SUCCESS in 1h 26m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b55961f952a479fbe7751881b17c466 : SUCCESS in 45m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e1be5b4a2bc47cab9ca6e5dc7eb3285 : SUCCESS in 17m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c2010f68a8a14b929385faecb9e11a01 : SUCCESS in 18m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b6f301ce22ad479792c47b67c52c53ba : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/50297472b6494a488c0650f9958e567e : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f27d71e273e8480e820415ddb8978974 : SUCCESS in 6m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7f1bab2e179b4cb088cafdd01acc32ce : SUCCESS in 7m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5e6d5ac6fe81408d928898298371fdbd : FAILURE in 34m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad04e035ecfb422e96e54b1be9451bc8 : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd03fb0d3cc04aec9b1c9a66af6cb8a4 : SUCCESS in 2h 01m 38s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/21416e84d85440ce912b76b1096f675b : SUCCESS in 1h 35m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b31ce2d407f540c7a508952e5210dc2d : FAILURE in 37m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/08c4dda0be244025b20084d7964027a8 : SUCCESS in 21m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fd65eabb2c4643e09d2d285e5123b60f : FAILURE in 43m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/86b9850184fd41c39ff3036151f817e2 : SUCCESS in 1h 00m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e85ac1e6b8874c52af5e011378473660 : SUCCESS in 33m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/af7cd6503ece41dbaf7dc3c0ecdc8b74 : SUCCESS in 1h 54m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/19292e35389e461091c30728de6392d5 : FAILURE in 35m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6298db148cbb446ebe16982e6779be24 : SUCCESS in 50m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f30bd4d7063e26d6e4a8051626cde98fa2953b0c","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-11-11 04:13:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5e49487e46032f5fa9cbb39d6b69e17f28d13164","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 04:36:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1f308683bd14dcc9108d3e1679cca41\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80b80be288f74a98a9533c4967d1d0f4 : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7bca5c4aaa69405db9bc40162f8466a2 : SUCCESS in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18286c4ec7861269d2a5ce1605d1bf27043b0641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 06:07:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/958f8ec3698b413ba83a7ce2afb3bdc8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18a622e24fb94a3e9738243b52a4bae8 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f5ac11e42dc45d3ac65d2628ebcd8d7 : SUCCESS in 12m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/320cf87cb8a74caf992996fccbf6a0e5 : SUCCESS in 11m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6224ab3616264bcdbd978be26296d8b5 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776cd8627a284ff8b376e1e73a7813e0 : SUCCESS in 12m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/c1f0225b7c4a45888a87e8f16e53369a : SUCCESS in 43m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9edca57ff5f45d8a0c162e12f5b6710 : SUCCESS in 1h 01m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ada15a26cbf445fb3b1774dca88fcfb : SUCCESS in 58m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbc563f54c1a4bcfbe1a466c35f36734 : SUCCESS in 16m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e80f737396974e9abccff7fda6bafc4b : SUCCESS in 17m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e2a0474b189140089daa239f1fe82fb1 : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cfb624be635f491bb25dd14c60e3211d : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1dbc20ba821b4009843a5aea53ed1bad : SUCCESS in 5m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0aa3764783a74336aca9438ca6183e01 : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ba9226b75f354328903c3cf3b65aed00 : FAILURE in 34m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0109c2949b9e4dc0b61d263bbcf8d4b1 : SUCCESS in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7a3d4071dd044718bc04340956c638e2 : SUCCESS in 1h 39m 07s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/67ff560659764f55838804f2c6af3a90 : SUCCESS in 1h 22m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/67164d33f23a43b9b6b313a380672664 : FAILURE in 24m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/24314d05bef648d29eba08f1454e912e : SUCCESS in 25m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e853dfe2cac44e4eacb867cde8f0fee2 : FAILURE in 45m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ccbadff6986748d4ac324c60eb0042b3 : SUCCESS in 59m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c83de703cc004ae9b701a60c01dd4bf8 : SUCCESS in 33m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7eff6b58675c46d189eac0ce8e4dadf3 : SUCCESS in 1h 48m 23s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4a716bf02cb847f297d267be1d555878 : FAILURE in 33m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09cebdf10d3b475aa16310bc91966777 : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":2},{"id":"96b8b0546b080a5cd114b78d1502dc4b04ec9862","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-11-12 05:06:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3cee893a0c22737db7c7b6e25e46201ae9426c63","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-12 05:14:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"d7fdf4cd76336e216a42c7246798b2397d9d1a00","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 05:15:29.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2522/.","accounts_in_message":[],"_revision_number":3},{"id":"cc43424bd08e47e997fe3964308a70050663b680","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 05:15:29.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2408/.","accounts_in_message":[],"_revision_number":3},{"id":"d2fa5ccd13731834c136523db129a8714d784fee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 05:25:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d0c06cf99964dd5b35bbd5b44f9960e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/babd7aeb32df45d1b91de5d3a8a1536f : SUCCESS in 13m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d1faa35f9f3f46e3a61825abcc0bd2d6 : SUCCESS in 17m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6b03c4ffd3c9d4d9c7ff085f108fb19d7ab44abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 07:11:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c593baea4d894f8fb7689dd742168877\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aa255a8ff0744a0be038b6ac7ce932c : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbbd63b204484402bc357d53fcf20c11 : SUCCESS in 14m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/334a104ca4364476846f3ca17acdb0f5 : SUCCESS in 11m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d98afcf803247d182032ef1750c3270 : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164e3dc6304b49e0bacfb611263a462b : SUCCESS in 12m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/5cd561df03a043fe9a68e400abbcd990 : SUCCESS in 55m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2084ea934f7348df80da75ed351f5d2a : SUCCESS in 1h 15m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32b4fde6c8ea4cef8897d7dbc411c8fe : SUCCESS in 1h 02m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68b1b58946de4190953c097f6665d429 : SUCCESS in 16m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fc1824e308b44da2a7d42cc8a463f6ae : SUCCESS in 18m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f5adefc9b5904f6cb6e3b634b08ed339 : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/71e1c23ee5fb4ea1b9544c2946cd1a1e : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4cce7277c9db4293b684faf65ad237c8 : SUCCESS in 9m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fdbdb5f6a97f4719af0b4b52bed3ae7a : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/34ef1f7bdab44900bee04caed9999aa2 : FAILURE in 23m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4fbbc2282d8148cd9ba0d087dfdf5d2d : SUCCESS in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f8c3cdfecf114339bf1679efaa33b8e5 : SUCCESS in 1h 16m 43s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a55ec47999834ff0ba68e064098b9cfd : SUCCESS in 1h 57m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/36c7c73d47f0433ea5229d911ac160cb : FAILURE in 37m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7cc41de66e9c4faf8458ace0a7dd0b75 : SUCCESS in 32m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/45c3350dc5114d2bb0ac32d1a70f1eb5 : FAILURE in 44m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0fa87fbb3922441c9031d0c2ced98301 : SUCCESS in 1h 02m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/81dca09ccc0c46f49453329c3bbe5378 : SUCCESS in 24m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fc69ae7b8c7a4980baf9573d7de25dbd : SUCCESS in 1h 11m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/930334d31dc344a6baa87fd4e1ca1e26 : FAILURE in 35m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2d2fb0f17974ee1985bbebd50d6cce4 : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":3},{"id":"2c4d41a53aa898907868de31034c20a712e1f345","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-12 09:51:55.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/964558 : FAILURE in 36m 47s","accounts_in_message":[],"_revision_number":3},{"id":"c561e586c40cfb36883d428dd9bf8cd1426ceb96","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-12 17:21:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e6fc7485cea1fc0d0ba183ef72ba25368c266711","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-11-13 11:12:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"847bf9c35bda09ef078c99ab4dd5afe7b60e228a","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-13 11:27:18.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_13915\u003e.","accounts_in_message":[{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}],"_revision_number":3},{"id":"ccde75b0851b143c32f1f9e21eae1f10e0740cec","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-11-24 04:38:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fbe284576a292296dfbfbeaf5c8e55f9fa516c13","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-12-04 05:25:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2d9c538d6186311ac5579071d179a3389bdd7a6d","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-12-05 14:20:24.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"49c4eb7a95b8fd1c24e38e2efe1bb68c0fb5a0c4","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-12-09 05:12:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d1db3ed88f72840d0b56d6c9f924eaeb762bb315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 07:03:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8adba11f3243451f8dcdeabc16a5ae01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8617523d7bbc49a6b538b40ac86cc1a2 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e91d8eb8236407a95751687cb942b60 : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/559cae2f3d91474788859863e69e70f2 : SUCCESS in 6m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04f7bc4370de489da007b460aa67e4ad : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f041d509bc745f7a05945a4f679ac39 : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/befb856827874025819520391c0651b7 : SUCCESS in 39m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d497b6b6f98b44cca858f20bdee68e51 : SUCCESS in 56m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58a90a439cfc4be388605023bb0d64ba : SUCCESS in 1h 00m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f53390f2753647df9d3ae6769a6c6fb4 : SUCCESS in 11m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d572fb0458514c2aa8a2c8a1377ed303 : SUCCESS in 18m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b6931d87b55746fe8f81bd960c451a7a : SUCCESS in 2m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/90e3ec66dc0e47fea94efc83b040da40 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9f0df5a698a9495cb69e4cc6d092b37a : SUCCESS in 5m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8591c04af592424ab61b4f4cae7cd88b : SUCCESS in 6m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f1453830fb643b89cc13f9749f30ef5 : FAILURE in 24m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6976920e6b934b7ead03ee8eab32ba78 : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/10cc9c52b0a54b8ea6b9613a1137ad09 : SUCCESS in 1h 45m 56s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3753887d64b846889d5c70cbd7400057 : SUCCESS in 1h 24m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f6a1f3f39ea74636ac26b74cae956908 : FAILURE in 37m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f92783e4e4344c40a250c58bae5ce180 : SUCCESS in 20m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b1268fe363d641789fe4501b6ac62abe : SUCCESS in 1h 08m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e3e0c87dbcd84e0b8d4ae6fe599c485e : SUCCESS in 1h 07m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0b09b5d04b5a416bbba04118845485e4 : SUCCESS in 21m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/61cefe3599d94a9c8e1eac8a9a5793db : SUCCESS in 1h 45m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ca145c41e9804f7886022fcfb9a919d1 : FAILURE in 15m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7eebab8cd9334d2aa09d49c2b8725486 : SUCCESS in 55m 08s","accounts_in_message":[],"_revision_number":3},{"id":"20c058f941efcceae071c0002e695da1b94b35a7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-09 07:10:16.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_12_2025_07_03_46\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"ecd2ab82562dd04e02c1d2e855be19b4f4eeabc3","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2025-12-09 07:17:30.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5328fb20a5330e2378680d4517b3b142c5f09f91","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-12-09 09:37:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70a4f24184413f213513e700d49357357db12c5d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-12-09 14:02:19.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1389419ee54231bda664934a96726c60f3ce1c28","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-12-09 15:42:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d04ad4d4b8a3a7c0f3f51ac280b4f95890b6caee","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-12-12 16:37:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e40e263054947e607f688a5ff328ab0517da3f1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 16:37:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7f31731f7282e4c088a7cc1167d37edb04a3bd57","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-12-12 16:43:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5e06301c46d46a0f916e4dad637281bd9efa1c28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:08:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cddd3c51cc334fcb98bed9ee45cc6a50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d0c822202e4d1a90d08d709072d5a8 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60eda43f6dc9429ab6c2b67c3ec899ea : SUCCESS in 13m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5ef0f42e8524c4cbfe5e8c39a03efd6 : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b74cd403c7e4f879307b4273e7c1842 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/818708a28d164366a4271d886864223c : SUCCESS in 51m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da73cb0716694c9eb96a4addd2334b7b : SUCCESS in 1h 25m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bca9f7e85ed4743ba5b85650f814924 : SUCCESS in 1h 00m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1192d8018964b2e817f5fdfd229c722 : SUCCESS in 18m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f6be20d72d464f42aaf85bc57b5dcf9a : SUCCESS in 1h 09m 12s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9c1287cef5c04479a52858fd1b480141 : SUCCESS in 1h 02m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0a53af380e544b0a27b29b3d7f0fe90 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":3},{"id":"7830e9dc25dcfd3470b119cfe5cda46a62133b28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:08:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f04f1ecf43fb403c053c374db17cad6b5c2e700e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:09:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9604eb723c4d405395a23545a9f8476e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b0bd4ec83ee46c68b2e8249f6e4d9a9 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1b45e6b397f48e49be7cc13ec59f62f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3},{"id":"1a714fe9df259296b7bf7a8593e200072181c296","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-12 22:10:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-964558-3-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-964558-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"f98dcb4a436745f05b5ab792dcb0361ab1875fc2","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2025-12-13 08:34:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"90adabb6771825506a9f860d5fc3fb21f5258f50","revisions":{"466b916223ff93543f754eb9a7a9b80fd424ba63":{"kind":"REWORK","_number":1,"created":"2025-10-22 16:12:24.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/58/964558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/964558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/964558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/964558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/964558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/964558/1"}}},"commit":{"parents":[{"commit":"b30d2f6fb5319412f3008942fec5b00a02beee88","subject":"Merge \"[Pure Storage] Fix issue with LACP ports not being identified\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b30d2f6fb5319412f3008942fec5b00a02beee88"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2025-08-05 12:34:58.000000000","tz":0},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2025-10-22 16:12:24.000000000","tz":0},"subject":"HPE 3par - skip license check for new wsapi","message":"HPE 3par - skip license check for new wsapi\n\nFor new wsapi version (of 2025), all licenses are enabled by default.\n\nThe wsapi does not explicitly list all erstwhile licenses (like Thin\nProvisioning, Priority Optimization, Remote Copy, etc) in wsapi output.\n\nThus, minor code changes required in driver code to skip license check.\n\nCloses-Bug: #2119709\nChange-Id: I9db121ac8ea631ae2043461929af0ee02ca05e85\nSigned-off-by: raghavendrat \u003craghavendra-uddhav.tilay@hpe.com\u003e\n(cherry picked from commit 94cd0e815fa3b00e3546a1986a9ce481749353b5)\n(cherry picked from commit 93d36f446755d2aa40043671cdb0d3687418a5ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/466b916223ff93543f754eb9a7a9b80fd424ba63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/466b916223ff93543f754eb9a7a9b80fd424ba63"}]},"branch":"refs/heads/stable/2024.2"},"489520c6e29ee0c11d239b5dacf5d164116f1fb1":{"kind":"REWORK","_number":2,"created":"2025-11-10 09:49:10.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/58/964558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/964558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/964558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/964558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/964558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/964558/2"}}},"commit":{"parents":[{"commit":"b30d2f6fb5319412f3008942fec5b00a02beee88","subject":"Merge \"[Pure Storage] Fix issue with LACP ports not being identified\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b30d2f6fb5319412f3008942fec5b00a02beee88"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2025-08-05 12:34:58.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2025-11-10 09:52:02.000000000","tz":0},"subject":"HPE 3par - skip license check for new wsapi","message":"HPE 3par - skip license check for new wsapi\n\nFor new wsapi version (of 2025), all licenses are enabled by default.\n\nThe wsapi does not explicitly list all erstwhile licenses (like Thin\nProvisioning, Priority Optimization, Remote Copy, etc) in wsapi output.\n\nThus, minor code changes required in driver code to skip license check.\n\nCloses-Bug: #2119709\nChange-Id: I9db121ac8ea631ae2043461929af0ee02ca05e85\nSigned-off-by: raghavendrat \u003craghavendra-uddhav.tilay@hpe.com\u003e\n(cherry picked from commit 94cd0e815fa3b00e3546a1986a9ce481749353b5)\n(cherry picked from commit 93d36f446755d2aa40043671cdb0d3687418a5ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/489520c6e29ee0c11d239b5dacf5d164116f1fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/489520c6e29ee0c11d239b5dacf5d164116f1fb1"}]},"branch":"refs/heads/stable/2024.2"},"90adabb6771825506a9f860d5fc3fb21f5258f50":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-11-12 05:06:36.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/58/964558/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/964558/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/964558/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/964558/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/964558/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/964558/3"}}},"commit":{"parents":[{"commit":"a54ed55cfbf1b38b09985058ccfa46a237d0ffd6","subject":"Merge \"zuul: cinder-plugin-ceph-tempest: raise swap size\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a54ed55cfbf1b38b09985058ccfa46a237d0ffd6"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2025-08-05 12:34:58.000000000","tz":0},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2025-11-12 05:06:36.000000000","tz":0},"subject":"HPE 3par - skip license check for new wsapi","message":"HPE 3par - skip license check for new wsapi\n\nFor new wsapi version (of 2025), all licenses are enabled by default.\n\nThe wsapi does not explicitly list all erstwhile licenses (like Thin\nProvisioning, Priority Optimization, Remote Copy, etc) in wsapi output.\n\nThus, minor code changes required in driver code to skip license check.\n\nCloses-Bug: #2119709\nChange-Id: I9db121ac8ea631ae2043461929af0ee02ca05e85\nSigned-off-by: raghavendrat \u003craghavendra-uddhav.tilay@hpe.com\u003e\n(cherry picked from commit 94cd0e815fa3b00e3546a1986a9ce481749353b5)\n(cherry picked from commit 93d36f446755d2aa40043671cdb0d3687418a5ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/90adabb6771825506a9f860d5fc3fb21f5258f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/90adabb6771825506a9f860d5fc3fb21f5258f50"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
