)]}'
{"id":"openstack%2Fcinder~965241","triplet_id":"openstack%2Fcinder~master~Id5dd417c06689ec9b5cea7f803b0a3fa057bae35","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"Id5dd417c06689ec9b5cea7f803b0a3fa057bae35","subject":"fix: resize restored backups to dest volume size","status":"NEW","created":"2025-10-29 14:37:39.000000000","updated":"2025-10-29 16:52:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"6a76a6d222bf561c002aacca3c8f437a9aa3527e","_number":965241,"virtual_id_number":965241,"owner":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-29 16:52:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-10-29 14:39:28.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-29 15:38:44.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-29 15:44:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-29 15:53:18.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-29 15:56:36.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-10-29 16:28:31.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-10-29 16:52:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27e70fe84356e09d60fc2a3be10ae10d171f4683","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-10-29 14:37:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"651421476249c7277d5ba3030ecb1f836c333609","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-10-29 14:38:02.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"42517f986a0c1d1784b3d9e1985d7b234e047288","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-29 14:39:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"0b9903588d2d291fce46c447aacde3a5a803362c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-29 15:38:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id5dd417c06689ec9b5cea7f803b0a3fa057bae35 .","accounts_in_message":[],"_revision_number":1},{"id":"f5c855dc97088de76e7810ef1c54ba2566357a83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 15:44:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f00a064bcf34dd2b68a7aa9b47ffb67\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e6d17f83507412ba03d58bc1309ae69 : TIMED_OUT in 1h 00m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7d6e7d94774b4ca7bb2bcc0c27bcb26d : TIMED_OUT in 1h 01m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5ad2d467afbb4cb5a17a3424ada04dfa : TIMED_OUT in 1h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fd8025b2b74c4408813c8ebe2044dd460a31f2a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-29 15:53:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1e9cdef05703420bb555cd2cef06af12 : SUCCESS in 46m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/edec92cd88ee446ab071db36c4290e2c : FAILURE in 58m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ffce71c0721be41dac52991fb911e6cedda9baf1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-29 15:56:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2200/.","accounts_in_message":[],"_revision_number":1},{"id":"82eacda77a6ea4a405bca2841e1ad50f17db4e72","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-29 16:08:47.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2314/.","accounts_in_message":[],"_revision_number":1},{"id":"7907c2dd8d7e3bdda17a96df309193c54bb49963","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-29 16:28:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/965241 : FAILURE in 1h 11m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6a76a6d222bf561c002aacca3c8f437a9aa3527e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 16:52:19.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f083b359daaf406492618d5fafbdfaa2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c1be760152640d9a9cdf392ff9226e0 : FAILURE in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a6dda7e48294a60b29a48d18028f1e8 : TIMED_OUT in 41m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39d2149f5fff45728bb7204a84bfaea6 : FAILURE in 16m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64fe7315e67f4441b6975b3141bc8e4f : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/27041b6f0a104bbdab01db5229d4f8c1 : SUCCESS in 56m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f73cfbee2e4e4a8db24541c4067979f2 : SUCCESS in 40m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5317c3966df4aa0a47db64ecb141b1a : SUCCESS in 1h 31m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fba055af87a4fe2a541828659df6605 : SUCCESS in 54m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/842ca931b0dc4132b698806daf08642c : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b69c9c4de5db48cd80edd199c3a7f00a : SUCCESS in 7m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9f256b1c44794061b1b1e495b5df6af6 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bd2492d4047840749a0c86a72dfd35f4 : SUCCESS in 5m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/67ee3bede1f74684b8500cbc01340928 : SUCCESS in 7m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e5fd567aee0f4b37ba1b551cd6288246 : FAILURE in 42m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/23979f7cf87b4eb6813d9aa7575c208a : SUCCESS in 7m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/04eb2ead98364410b67591fdc9acdf4f : SUCCESS in 1h 22m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/658282a8fc01497ba4188965026ae62d : FAILURE in 1h 35m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/24ff41b92aff4271a2c36d9574b0e312 : SUCCESS in 1h 59m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/60230580f5e8458db5af8880904c6c1a : SUCCESS in 1h 53m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ebd017c1f1f64ce1bdadc43e5e8f4af3 : SUCCESS in 34m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c3eeb265197648088faea1d1ff129f47 : SUCCESS in 1h 12m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/20611af330e447bf89ebd26fe90fb3aa : SUCCESS in 1h 24m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e459847708f148e782273292226ae94b : SUCCESS in 32m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b4ec2de2bf74d6a9fa177be681df681 : SUCCESS in 1h 55m 15s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bde25cb5b7744a7fb96821b2580da7ee : SUCCESS in 2h 06m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7539844a457944b690d4a176906a4f37 : SUCCESS in 1h 39m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ed5dae620ee4d5880b2fb3ed3abad5f : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fee19ac8ce35cc019793017ed18211833771c6eb","revisions":{"fee19ac8ce35cc019793017ed18211833771c6eb":{"kind":"REWORK","_number":1,"created":"2025-10-29 14:37:39.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/41/965241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/41/965241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/965241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/965241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/965241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/41/965241/1"}}},"commit":{"parents":[{"commit":"d02171164bdd702b12b59888b744d172f30d712d","subject":"Merge \"docs: starting iscsid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d02171164bdd702b12b59888b744d172f30d712d"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2025-08-27 17:20:44.000000000","tz":120},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2025-10-29 14:37:05.000000000","tz":60},"subject":"fix: resize restored backups to dest volume size","message":"fix: resize restored backups to dest volume size\n\nWithout this patch, when restoring a backup to a bigger volume, the\nrestored volume will have the size of the backup and not the size of the\nvolume. This can lead to issues when the volume metadata has a bigger\nsize than it actually has and also the volume/image inspection might fail.\n\nThis commit adds a resize operation after the restore to ensure that the\nvolume has the correct size.\n\nThis also fixes restore consistency due to the skip of restoring zero byte\nchunks. see if condition in [1]\nIf the source backup has no data in the tail end of the volume,\nthose bytes will not be written to the volume and the size would be less\nthan expected.\n\n[1] https://github.com/openstack/cinder/blob/d02171164bdd702b12b59888b744d172f30d712d/cinder/backup/chunkeddriver.py#L78\n\nCloses-Bug: #2130224\nSigned-off-by: Max Lamprecht \u003cmax.lamprecht@stackit.cloud\u003e\nChange-Id: Id5dd417c06689ec9b5cea7f803b0a3fa057bae35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fee19ac8ce35cc019793017ed18211833771c6eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fee19ac8ce35cc019793017ed18211833771c6eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
