)]}'
{"id":"openstack%2Fcinder~965837","triplet_id":"openstack%2Fcinder~master~Id91b76312551dd8ee19f452b3d739063eaa0b050","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-12-01 21:52:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id91b76312551dd8ee19f452b3d739063eaa0b050","subject":"tests: Add API sample tests for os-services API","status":"MERGED","created":"2025-10-31 17:40:18.000000000","updated":"2025-12-01 21:54:07.000000000","submitted":"2025-12-01 21:52:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"965837-openapi","meta_rev_id":"f47044fb2f1f9a0dd0a42a9409e319d2336e7ee5","_number":965837,"virtual_id_number":965837,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-01 21:52:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-11-20 21:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-12-01 20:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-12-01 20:05:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-11-20 21:58:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-31 17:44:42.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 18:42:27.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 19:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 20:02:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-01 12:36:53.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-01 13:26:01.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-02 01:52:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-03 20:16:42.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-11-20 21:58:21.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-12-01 20:05:01.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-12-01 20:05:17.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"}],"messages":[{"id":"6dda95bde6551cf1295a9fc96d1bb7dabf542d1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 17:40:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"066b3a7c2a328f3b349a33340e8f4e0ca83fe534","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-31 17:44:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"6a895644986995a955fd8a0a579b88a9c035a882","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-31 18:42:27.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id91b76312551dd8ee19f452b3d739063eaa0b050 .","accounts_in_message":[],"_revision_number":1},{"id":"2ad73d225fbc0cda888384e3fac374da76c86a10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 19:39:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d60b86c76884b78b2e075a46d8ad777\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ac3e1a19a4f14a0eb1af034e4774dd9e : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d8c60b3f8f054150b5b3c3804c00381e : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94fd658706ce4d4ca26c15d10538ea8b : SUCCESS in 19m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fd3c634cded526a1c1bb7a1794470b19c1aae51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 20:02:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7567312d691e46649b899745792ce607\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/98dc902535bb4ef7a5a3da00c4ef2bde : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02435c612b864b9dbb21778dbf6a4b08 : FAILURE in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52e4f9aaf8f04e6fabd8ee3eb55ec1d7 : SUCCESS in 12m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/940a7e74f08648098f962b28360e6183 : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98180e35869a4c5daca596673e3a74ff : SUCCESS in 14m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/98770c217455416ebd03cab05b9e3d12 : SUCCESS in 1h 01m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b3d03623732c4ac781da1f9ee33ddb6b : SUCCESS in 1h 05m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/176980a809d84842bb0fe35de61b8cb5 : SUCCESS in 1h 33m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea880d383cf04e118f861bb0f780b957 : SUCCESS in 1h 04m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e9b13735823e4405823c23c9797b726b : TIMED_OUT in 41m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/48f5a964ba51471a9d9fa176d8932a0e : SUCCESS in 5m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/91623bb8cb3a4682b87d06e764fcb276 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1fe7816173a84650a4abe6d4a3289228 : SUCCESS in 6m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d0465a651e994ba1a68419bbac2eeec5 : SUCCESS in 10m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/21be8597df2e41108caf8e1332f71f63 : FAILURE in 39m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fc003fce7e1448d885219e2870284012 : SUCCESS in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dc3095a000284690a85c28bf09aafed3 : SUCCESS in 1h 50m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0cc28c9fb20846a1bd48a687e7890489 : FAILURE in 1h 48m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c82809ad8cd54a8fb4611900a946f905 : SUCCESS in 2h 14m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cfbfdda4750642fd9a5419157a64b85c : SUCCESS in 1h 44m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d2536abb946342b69b96aa276cbc4be5 : SUCCESS in 17m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c420d8983ea44968bfbf7180b622af05 : SUCCESS in 38m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b0067cf736a0441c8d26c819913e0d80 : SUCCESS in 1h 40m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7561100edc7f450f8e6ab85c78ab9cea : SUCCESS in 34m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/56c8db6468d540f2ad3b88494f5fbc7f : SUCCESS in 1h 32m 01s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/991c7334786744399803628571a3c5df : SUCCESS in 2h 13m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a3b746eb44649cbac44f64f39177683 : SUCCESS in 52m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64982b9f2c63420a941cb87beb8815e5 : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":1},{"id":"48f5fd06b0350701197f884c1266e0c62618572d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-01 12:36:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/965837 : FAILURE in 1h 11m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4b6490b3835bc7f04b83abdda600cf6e70f814eb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-01 13:26:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2268/.","accounts_in_message":[],"_revision_number":1},{"id":"5ad352a7669e4decde12725528d3764c705a14b7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-01 15:02:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2382/.","accounts_in_message":[],"_revision_number":1},{"id":"ae88b918a91d482ab96ff489c5d959eddfc69308","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-02 01:52:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0bda73d8e4d74ced95b066a457b51d32 : SUCCESS in 1h 37m 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b1dfa74c06724f628032588bd07acdb3 : FAILURE in 1h 53m 02s","accounts_in_message":[],"_revision_number":1},{"id":"16804497394cb6f405f107ddd142bd55d5514e00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-03 17:28:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a4e1019f7f8db4495aefc4b17fed680a3fb5133d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-03 17:30:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"968b3937637b2040e233aea62fdd719267b666c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 17:59:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f3d5c5d5a574a56a08ed7cfcf80b3d6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c19af5c7c04c4b8391adafd8789e71b4 : SUCCESS in 22m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f041ecc6c140461884e96573d1928b72 : SUCCESS in 22m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2dfdb4d6a82948ef9c3e97b3c6df5f4f : SUCCESS in 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"697f419442ba90578093723ddc2aa61ee888cd19","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-03 18:33:31.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id91b76312551dd8ee19f452b3d739063eaa0b050 .","accounts_in_message":[],"_revision_number":2},{"id":"59f1eca0b1609f7582e3ba0599077f18844cbc1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 20:09:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eaa82fcaf9f409baf2a496a3f6f1599\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c80a69bcc44f4b87b0c3b50625c41d3f : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abea0045d5134a35ad090ee3405c3ecd : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32d99324e3954632852748351f14785c : SUCCESS in 9m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30b6ca9d5ed340a89ea6f2f3b0f3e539 : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e006d19fc78445fe922ebd43e1ece6f4 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/1fa7f6c02c71483d8d172d23eaee82f3 : SUCCESS in 58m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/51fca6b0695a4850adbdb6d81d635a18 : SUCCESS in 46m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91d1d1e8827a454cbed484d94b615482 : SUCCESS in 1h 35m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d36a2c9e42bd4c24823d6a7725b671ac : SUCCESS in 1h 21m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7cea53161e5e4d8696d5d6f3efeee8fe : SUCCESS in 21m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/556bec38495b4822a307dcd24602ced1 : SUCCESS in 4m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2b08c5e70bf04a469748ffcb29fc7904 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/411052680a314753929e8a2ba50f9624 : SUCCESS in 5m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/92510821842b455695ea362cb5ef5278 : SUCCESS in 8m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/336812459856472ab934c1fc7c9b993d : FAILURE in 56m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/86dd3fdb9aa343bca260dfce9eaf58ea : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/35438eccb5c84c0b8691cec2b56dd777 : SUCCESS in 2h 01m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c95b3ca77a014f84aeacfe7b41a67b31 : FAILURE in 2h 35m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8942fb559a204b89961efc52100cfd11 : SUCCESS in 2h 08m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cf988d1859574c4784b92f9b5d8ca12e : SUCCESS in 1h 39m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5103478132494103bc83aa14637487c7 : SUCCESS in 35m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7d0917df6a434df9b18d7a623f6c611e : SUCCESS in 1h 08m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8cd8b6840d6a4578baa15c7653833ed2 : SUCCESS in 1h 39m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/77f6317b5c1041e68c2adaa9e4b5ba1c : SUCCESS in 34m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/45f4bbe895044c6ea1fd8832404930d5 : SUCCESS in 2h 02m 51s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9d10ef2046e445cc916128dd822cb73b : SUCCESS in 1h 23m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/149dfa2254694467a347e79759270212 : SUCCESS in 2h 01m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5a6df005e214c34920928dec029babe : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":2},{"id":"00ad898897fc89983b3ec047560f8eacb84ca03d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-11-03 20:16:42.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_11_2025_20_09_57\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"7fd3a6b0be4c15c93987e26ee92e01e8a1bc0aae","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-03 21:40:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/965837 : FAILURE in 1h 12m 45s","accounts_in_message":[],"_revision_number":2},{"id":"05f4aa30ccb2e57d6870f21380c7a814a47b4bbe","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-04 00:58:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8a3790b0bc70403599db9e8067537e22 : SUCCESS in 1h 38m 43s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e732a8fb8c7642ff9c6240484289b555 : FAILURE in 2h 18m 32s","accounts_in_message":[],"_revision_number":2},{"id":"bdb0254bc47d6bf523856660ec2c6cee9e0e9e4f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-20 21:58:21.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b976fdf82083384be44df21d1f2b9d847539cb64","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-12-01 20:05:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"62f0e3511fc6e8df46ef0eef0232d63cf15b5f1e","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-12-01 20:05:17.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/965836 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"2aad709422bd880b9752dcc4c69a091639b55a21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 20:05:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4722495dad4f38eb73c7325ddea3cd0c3a20e99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 21:52:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae893ed6fa5d444086bac19160d7b80d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aedc72e3024b4cd9ab8e41792066d5c3 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f64dfaac9fc4f768518b179b2a26d1e : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34ada211c04842c9bcac7eda8633713c : SUCCESS in 9m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f960148b93fe44f6beffdd86fd37a001 : SUCCESS in 15m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99d561e113347fe820830814329d252 : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/7e1a3b9b7d8847a3bb8923511217ab84 : SUCCESS in 28m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d2cfbf01606c4c1f9d72a4cc45f89696 : SUCCESS in 28m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/feda0841984042b28d5893b6bfedc4a5 : SUCCESS in 1h 21m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa2f1b3c92ca45ce9a23cd168de0af55 : SUCCESS in 59m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3238c51a293b4430b4cb5a19ace4b146 : SUCCESS in 1h 09m 51s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd0c8b7294c94703be4d1d908973f317 : SUCCESS in 1h 32m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09ce6457c91647e0833fd1f8afb44f8c : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":2},{"id":"abfb587531d357d0d71f1926f83ab475ba4a2c35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 21:52:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f47044fb2f1f9a0dd0a42a9409e319d2336e7ee5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 21:54:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/964e62b361064ca08586f3e2c445a852\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/704523692c4d4e51b2b95e807820de84 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fba90d582e9940939830595674634f2d : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ea35f8374241db307d2e4db075a1c6a3a0a08af","revisions":{"b30cd8035d0ec20817e15f6c2bc74c137e37c8e4":{"kind":"REWORK","_number":1,"created":"2025-10-31 17:40:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/37/965837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/965837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/965837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/965837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/965837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/965837/1"}}},"commit":{"parents":[{"commit":"91b5501585544cadeed9d569556332b80c4e0c56","subject":"api: Simplify enable/disable APIs (clusters)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91b5501585544cadeed9d569556332b80c4e0c56"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 13:00:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-31 17:39:33.000000000","tz":0},"subject":"tests: Add API sample tests for os-services API","message":"tests: Add API sample tests for os-services API\n\nWe\u0027re going to be conducting surgery on this API shortly. Let\u0027s make\nsure we don\u0027t regress.\n\nChange-Id: Id91b76312551dd8ee19f452b3d739063eaa0b050\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b30cd8035d0ec20817e15f6c2bc74c137e37c8e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b30cd8035d0ec20817e15f6c2bc74c137e37c8e4"}]},"branch":"refs/heads/master"},"2ea35f8374241db307d2e4db075a1c6a3a0a08af":{"kind":"REWORK","_number":2,"created":"2025-11-03 17:28:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/37/965837/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/965837/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/965837/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/965837/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/965837/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/965837/2"}}},"commit":{"parents":[{"commit":"91b5501585544cadeed9d569556332b80c4e0c56","subject":"api: Simplify enable/disable APIs (clusters)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91b5501585544cadeed9d569556332b80c4e0c56"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 13:00:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-03 16:02:36.000000000","tz":0},"subject":"tests: Add API sample tests for os-services API","message":"tests: Add API sample tests for os-services API\n\nWe\u0027re going to be conducting surgery on this API shortly. Let\u0027s make\nsure we don\u0027t regress.\n\nChange-Id: Id91b76312551dd8ee19f452b3d739063eaa0b050\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ea35f8374241db307d2e4db075a1c6a3a0a08af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ea35f8374241db307d2e4db075a1c6a3a0a08af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
