)]}'
{"id":"openstack%2Fcinder~965840","triplet_id":"openstack%2Fcinder~master~Ia8b0fe29283742adfb345bb2b9f6f6b996791f7e","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-12-12 16:04:45.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-12-12 20:34:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8b0fe29283742adfb345bb2b9f6f6b996791f7e","subject":"api: Rename controllers","status":"MERGED","created":"2025-10-31 17:40:18.000000000","updated":"2025-12-14 20:10:30.000000000","submitted":"2025-12-12 20:34:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965840-openapi","meta_rev_id":"ce273687067420cc719f6d8ec4020f8c7dcc8caf","_number":965840,"virtual_id_number":965840,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-12 20:34:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-12-12 16:04:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-12-11 17:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-12-12 16:04:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-31 17:48:46.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 19:32:38.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 20:06:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-01 15:53:01.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-01 17:21:12.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-02 17:32:30.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-03 21:38:55.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-12-04 00:05:28.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-12-11 17:45:25.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-12-12 10:29:23.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-12-12 16:05:01.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2025-12-14 20:10:30.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"1a443ae4d06fe95cc94f6a0332cdd10bd9936027","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 17:40:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"babf7a94ada156016a66b48151b81a243d66b957","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-31 17:48:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"44d6302355aa7c0e728c4ed608e094c21fa6a40b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-31 19:32:38.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ia8b0fe29283742adfb345bb2b9f6f6b996791f7e .","accounts_in_message":[],"_revision_number":1},{"id":"e78e142682e006fe91eb109d1e72ce87c7c9984d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 20:06:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c430f5847b54c0d99c78a882ceb73c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72a4fae13b2e4b54a1044ef9f40e8f8e : FAILURE in 7m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13e5079d63824eccab53a55966960cb8 : SUCCESS in 12m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a6822b66ac54c1ab50fb0baba46832c : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679546357a5b4d089c3410bedb8a4ebf : SUCCESS in 15m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/94c8327b9b3c4de2acf7fe07b126d6eb : SUCCESS in 1h 03m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1791cdfabec048ba9184ef849bfbd07d : SUCCESS in 57m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c20166cf62924998982e9a2a47db6f96 : SUCCESS in 1h 46m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c8f26c2a55c49af8b4ff2baf914532b : SUCCESS in 1h 06m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/478add5d201b489eb856efd79850493d : SUCCESS in 18m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c29ef2edde8b402ea222802a10d18609 : SUCCESS in 5m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba7223b081154b81bfda4b13b13d1fb5 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d74a5b9bc2324921b02a1d9cb6aa8162 : FAILURE in 6m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dc5c009eca3d48918806e927afc3240a : FAILURE in 9m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a1265759e3ae4b3b878f1d4fafb1ac21 : FAILURE in 48m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5938cb4117c4fcd9ac3213721519d62 : SUCCESS in 7m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/36858bba81dd4a7aaa8f970628f3a11e : SUCCESS in 1h 51m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8401515c28c24460a4415aaa9eb5eead : FAILURE in 2h 14m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a9c84fbc99f942208cb55b08bc67964b : SUCCESS in 1h 45m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1e8e791f08a74182b450518dd6904a4d : SUCCESS in 1h 45m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/280853615b72445c83ae68f3656661f1 : SUCCESS in 20m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e26e07a9f31b4725906f939c1b7b9539 : SUCCESS in 39m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/877ab30717cf4261a112d9681d3e9fbf : SUCCESS in 1h 49m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bdaef97bac9d492483d52e3fe098e38e : SUCCESS in 44m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2b980c066b0d466d95a67d67f5622bd4 : SUCCESS in 1h 56m 54s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/27a0905baed645b997cadf2a4b34d92c : SUCCESS in 2h 18m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b52bcd2521f54faf92a679865798c589 : SUCCESS in 1h 45m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c52e740fe7c4aa8a03245a883b5bfb7 : SUCCESS in 57m 39s\n\nWarning:\n  Comments left for invalid file cinder/backup/rpcapi.py","accounts_in_message":[],"_revision_number":1},{"id":"c21628adee80851755a5d87f5f9eab32dd719cd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 20:10:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45ac45cdedc043e6843a92adf3c1c036\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c603f01edc0f44f5ae2c9775955b7deb : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84faaa7f565b4e37b582f78237bb7ef3 : SUCCESS in 21m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6d4b4543ab404382bf393dfb9e4285da : SUCCESS in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c9587e9a282fe875d83d8da3f367b31c0e355ee","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-01 15:53:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/965840 : FAILURE in 1h 10m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b630bac6e2d7df2e45a103847669fb8944243645","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-01 17:21:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2272/.","accounts_in_message":[],"_revision_number":1},{"id":"21bbb2b29534a73319ad01f79dbe5b8f18ad6039","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-01 18:31:32.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2386/.","accounts_in_message":[],"_revision_number":1},{"id":"6c22766d9f6322655cd9660d02dbb759e4953e11","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-02 17:32:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c8a91326e472495e947b6405d027466d : SUCCESS in 1h 37m 20s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4773da68d01d416c9c16b0f139a917e6 : FAILURE in 1h 49m 26s","accounts_in_message":[],"_revision_number":1},{"id":"5dec03478ea199339f4ff9d5819b67d2f624c816","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-03 17:28:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"901a19cca8a80969b5dd6d0def9ccb51d5d4cfbb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-03 17:33:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"818ae7acfdde6baad825402cf761a055ac310c73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 17:53:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c9dbbf4392e4c6fb6362048435c1b94\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1f120b7f38e64a95b5e92d58ec387c40 : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a76507cb46cc47c7bee6b59db47cdcac : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/308f55e78fdf4420b7541959e8869b38 : SUCCESS in 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0d455bdd36505a43da0bd9bead8e3b2db83b4d0a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-03 18:32:29.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ia8b0fe29283742adfb345bb2b9f6f6b996791f7e .","accounts_in_message":[],"_revision_number":2},{"id":"ab43dee3566ad438f2e99b071d8e18eefefad597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 20:52:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3593ce953ab24c25ac77fa8fef1cb491\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b9d8de03a8a4918bd6de973e31fb52e : SUCCESS in 8m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e5b01c0e4284123a94ba2c59ecf0ee7 : SUCCESS in 16m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6be6af701f2473bb1e666a231ac1305 : TIMED_OUT in 41m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97b5eadb0b434f68aa0e9679046db5f3 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/4f264977fb794f5ab4cfbd5660410165 : SUCCESS in 58m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bdd68c23c1524f6c95b04e49ca12f4e6 : SUCCESS in 55m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8629381654c043e9ab698a0250ecae75 : SUCCESS in 1h 30m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/548642f9da1d482dbdb783088239d8d3 : SUCCESS in 1h 06m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3ad825abe9d94002b4d3911aea4ddcb6 : FAILURE in 27m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/36fe50d148d645eb83478936c2aceb63 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c0b097e691114d799ea56b20990406a0 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/68604567133d4685a14eb5df015efe62 : SUCCESS in 6m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/88a2d3939165431b9918ae055093b91b : SUCCESS in 10m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1df17849c9bc4237a76e95f57b192ad0 : FAILURE in 38m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f65bfd700704e23b6b767e64c82a291 : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/83250f05211947159311f3587428cd8b : SUCCESS in 1h 34m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e51625018ea348a0b4821dbbdad2e92d : FAILURE in 2h 56m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ac69059c14af4eed964bc4f0e1c0853c : SUCCESS in 1h 30m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ab8cfd92cd974b3fba31c4702f4e90a5 : SUCCESS in 2h 10m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cd3ead4fcf294c309af8f8b7626cc543 : SUCCESS in 35m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d31cee4f3a84135a4f6880001c61b13 : SUCCESS in 43m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/77e9b3700942490793ce087289d28ccc : SUCCESS in 1h 44m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0a6efaa9cdb14611a8d5ebebf6964585 : SUCCESS in 47m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d00ee42987c940c9b2ed61814dbf9124 : SUCCESS in 1h 12m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/7a89634de63f491aaefee55d3cb94f08 : POST_FAILURE in 2h 39m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d26c0eff5834691a737fb93e8e89381 : SUCCESS in 1h 53m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adbcbff5852a4536a4d762b46ce72264 : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6a0d80c94f35a8d9fd0bd560f94c770ce4bba0eb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-04 00:49:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/965840 : FAILURE in 1h 10m 59s","accounts_in_message":[],"_revision_number":2},{"id":"ffd35fc7fe12bcdd9cdffbdedaa0ed879008eec7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-21 10:46:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48ed7dd70af1e2a94ac4b8ddf1e5820ba1c096b9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-21 12:48:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0f1e2ec59e054f43b649cd219b5ce232 : FAILURE in 46m 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dd738ea1989d4c798d3cd9e24b528ee7 : FAILURE in 1h 04m 14s","accounts_in_message":[],"_revision_number":2},{"id":"a60465dfaefc8b743df2d77850e967426561cf29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 14:10:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1eaaa3768ea04c08b637bdff506b540f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dad2da7aac7840358a374f00e6e25250 : SUCCESS in 8m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76c9dbf1e74246aba4ab0e05ba6bb0cb : SUCCESS in 12m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2ac76b55ac44631aa2a561a1f43e4da : SUCCESS in 14m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b59da8267294a428c3ba92839be839e : SUCCESS in 12m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/de18e5b010f347b98deffec1022db3be : SUCCESS in 1h 05m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85dd467d29224f8f90c4abccfb608914 : SUCCESS in 54m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3cc86df4a58b434cb46e080cf320800d : SUCCESS in 1h 50m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/246a170874a844fc99c40835b6085ed3 : SUCCESS in 1h 05m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1cc4fe960a07427ea00d83dcff511390 : SUCCESS in 21m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a7c9feaa097c4baca5bc16a8fcc8be52 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/621cfd7fac23496d8b635ea78fec52c9 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4e26e5e88e1544c5aa58ed612ec6a24c : SUCCESS in 6m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f8726222c7f4406495aa30adf0319954 : SUCCESS in 10m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a5b5d7bcbd0345e0b100e3686fcbdc45 : FAILURE in 40m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c6f190220bb4dfdaef5e06244e5259b : SUCCESS in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0ad95f383607417fa0ee48ca23c9645a : SUCCESS in 1h 52m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/feec973dfd7a4c608b101ea3fa674839 : FAILURE in 2h 54m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b9424795b8024a7c9e09fb0bfd3df11b : TIMED_OUT in 3h 08m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e0ac06a2a8944c8fad9c0b6303873b3d : FAILURE in 1h 48m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2d9d833fc2904973aab8c7f2ed274413 : SUCCESS in 28m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e2982bddfc4948579377ae16fc451a91 : SUCCESS in 49m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/983378be19394967a9236d8b98ea2dcd : SUCCESS in 1h 13m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/99fc1eb574e54cc9a55887d169b437c1 : SUCCESS in 38m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ca634d50957d44aa9a12af1161f1b4f0 : SUCCESS in 2h 14m 36s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/71519e2580944eb380d2bbe6a3b174d3 : SUCCESS in 2h 13m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b76174d0f09e4e11b09f344491a81724 : SUCCESS in 1h 50m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe028995a5a34d47b17b1015e1bda977 : SUCCESS in 44m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f602999b7443daaecac02dcbad37707b3eec223e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 14:47:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fe9dcf4342e4d2084f85a45d0215136\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a49a5b3b863142eb962ade29a2a7d458 : FAILURE in 32m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ff07478a4384561886eb429a9f6f81b : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/21a226cea4ee4f66afecd3bde618bd58 : FAILURE in 40m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"31e86c8b6ccb5ce33a55c316e9d0472d14f61b93","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-12-03 21:38:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7aa5be01a417fbd2a9a9067573b1ffc5d886eb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 22:02:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe224f74fa3140d182cdc1b793119c56\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/17e1162d312f4a8d8fc65fc5b5e97b8a : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9525825ded5142eda531b815e40a4a0b : SUCCESS in 21m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3fb079e896854fa89514378693926923 : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"98a79c2b4be6e32f3759106dcf25560ff46e6e23","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-03 22:36:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1a02ae74844744cca7a40f8887ecd791 : SUCCESS in 52m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/02b25941fc7a49cb9ad798e031128987 : FAILURE in 53m 14s","accounts_in_message":[],"_revision_number":2},{"id":"40f6ab4cd719a510e30e4a6943726a2377162b44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 23:58:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca125fcf3b8e48bb9b8c32bdc1f7d646\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ce625efa37494e90ebbcb685bf91fa : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce5b6b70c52443508e5e7155b87708ce : SUCCESS in 7m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68c4dfc55d6140719a4c41c0892b6694 : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12c7b4310c5a46ab927c903ff0738d5b : SUCCESS in 16m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/f8763e2598d245dd8e87e135ef87197b : SUCCESS in 33m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb9f4409e22e448b9fd2bdbbc49d03c1 : SUCCESS in 1h 02m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4af9275b2f844d1683cc9d6eac11d84e : SUCCESS in 1h 21m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32d791ec40ce4cf39f2e265b931accd1 : SUCCESS in 44m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d2bbdff05cd447d8ae3123de2d02adec : SUCCESS in 19m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8503daf7219046d5a70bebd0c4b9f660 : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2583c3033b48428aa0a3776902489c75 : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/904f2d02ea44484391a6c92380775ec3 : SUCCESS in 8m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b7f62aa08e6a4f4b9b4d5edbd7e41c8e : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/448877d3cf6743748aa8a731dcf648e6 : FAILURE in 17m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1916b27575b94f379cad6a0359c21da2 : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4b9fb1de15b3417bbc4541c8be43a43c : SUCCESS in 1h 54m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f3cfb8ded5e84038bcc7957728cdf932 : FAILURE in 2h 13m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/392c7f9087ae4879a285ce8860bc2689 : SUCCESS in 1h 43m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4567752395bb4fa481617c79d58131fd : SUCCESS in 1h 53m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5d58ef79d88a4afc8ff138acac8f4335 : SUCCESS in 34m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5e3a6de6407b49d88ff95bdbdebf16a1 : SUCCESS in 1h 07m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/946c0828dcb4493b9b0700a490c4f79c : SUCCESS in 1h 29m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/98ed5d0ef0aa4ce6988e44953a2db005 : SUCCESS in 43m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c1b5c15fc40c4b97847715a8f5ad5dd1 : SUCCESS in 1h 04m 38s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/59c5e29e248b43d9a532973d45f56ca2 : SUCCESS in 2h 03m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e61b020a845479b8d872256492aad39 : SUCCESS in 1h 46m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e58795465f26457095ecf1ca18d7b067 : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":2},{"id":"60d8dd320d61f5a8d8d4ffc73ba23fc86958f5d5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-04 00:05:28.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_12_2025_23_59_05\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"bb62f2517c455eaf2b240672915889296ac80ca9","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-12-11 17:45:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5727a32400ee0ad919146799942e15c715061543","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-12-12 16:04:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b33e3e2f53e50a68a66ddc5fd39f639f641cd35e","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-12-12 16:05:01.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/965839 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"2a36166c2f0024711abe139a7f6facf13e477667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:33:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"12343f342f3b2baf72b65bec9b96b63ca2aa7875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 20:34:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b325f694768d4d398d66bd64d9387711\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/233aae155d7940aa995ca9da73592ff1 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91a5526b2d29486c9f4ab46e62e023fa : SUCCESS in 8m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb0518c1c0474656b2604ea21e073905 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86649a6c71b84a51ae46e32b961fe846 : SUCCESS in 15m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/c34dbee6cb0047fead3d67609b71d227 : SUCCESS in 32m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/83bc632ca6954396a92d2a65b0cbf073 : SUCCESS in 33m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e477ed44276415eabfbce5fdd0e0010 : SUCCESS in 1h 18m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37f4ba81ae20467ebc9ce69c73e8150d : SUCCESS in 36m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2cb93c415d0347f48088d4f563b613bf : SUCCESS in 38m 53s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9ca2ef7381894da5a23936809124e04d : SUCCESS in 1h 31m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd11bc3bc527497c90e8e97c5712e1dd : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f585258fdecaa472337207a7453503ecb1f87867","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 20:34:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"394ca9ebc5574f41e066cf22f07c2d30858aa3bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 20:36:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e34654e7735b46159b75cf0593d3dad4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7d22d0671a4f248409baf36a652fdc : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2},{"id":"ce273687067420cc719f6d8ec4020f8c7dcc8caf","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-14 20:10:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-40-965840-2-fc : FAILURE in 1h 15m 36s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-40-965840-2-iscsi : FAILURE in 1h 11m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"75d12bc023e31250a1cabb87e6c1b69e03c7012a","revisions":{"c34bd8ef77a6e7baee6eb6cd5cc7108e08d7e115":{"kind":"REWORK","_number":1,"created":"2025-10-31 17:40:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/965840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/965840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/965840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/965840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/965840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/965840/1"}}},"commit":{"parents":[{"commit":"71b9cbe22f25455f4ff734691dc3aaa8b2178c63","subject":"api: Merge os-services extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71b9cbe22f25455f4ff734691dc3aaa8b2178c63"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 15:51:42.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-31 17:39:38.000000000","tz":0},"subject":"api: Rename controllers","message":"api: Rename controllers\n\nMake the logs for failing tests a little simpler.\n\nChange-Id: Ia8b0fe29283742adfb345bb2b9f6f6b996791f7e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c34bd8ef77a6e7baee6eb6cd5cc7108e08d7e115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c34bd8ef77a6e7baee6eb6cd5cc7108e08d7e115"}]},"branch":"refs/heads/master"},"75d12bc023e31250a1cabb87e6c1b69e03c7012a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-03 17:28:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/965840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/965840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/965840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/965840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/965840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/965840/2"}}},"commit":{"parents":[{"commit":"0fcc22942d603bd3c06114b97faec53b2f03ad1c","subject":"api: Merge os-services extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fcc22942d603bd3c06114b97faec53b2f03ad1c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-23 15:51:42.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-03 16:07:24.000000000","tz":0},"subject":"api: Rename controllers","message":"api: Rename controllers\n\nMake the logs for failing tests a little simpler.\n\nChange-Id: Ia8b0fe29283742adfb345bb2b9f6f6b996791f7e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75d12bc023e31250a1cabb87e6c1b69e03c7012a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75d12bc023e31250a1cabb87e6c1b69e03c7012a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
