)]}'
{"id":"openstack%2Fcinder~965841","triplet_id":"openstack%2Fcinder~master~I0d4ed51932ba94abb59e650fa726c877b94656b1","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-12-12 23:47:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0d4ed51932ba94abb59e650fa726c877b94656b1","subject":"api: Add remaining missing request body schemas","status":"MERGED","created":"2025-10-31 17:40:18.000000000","updated":"2025-12-13 21:08:48.000000000","submitted":"2025-12-12 23:47:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965841-openapi","meta_rev_id":"bea4235fd1383fd4c712ef6be1f36a6250538900","_number":965841,"virtual_id_number":965841,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-12 23:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-12-12 22:06:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-12-11 17:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-12-12 22:06:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34470,"name":"Yan Tseitlin","email":"yan@lightbitslabs.com","username":"yants"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-31 17:50:22.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 19:24:21.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 20:10:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 20:43:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-01 16:21:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-01 16:34:14.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-02 08:34:26.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-03 21:15:59.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-12-02 13:14:09.000000000","updated_by":{"_account_id":34470,"name":"Yan Tseitlin","email":"yan@lightbitslabs.com","username":"yants"},"reviewer":{"_account_id":34470,"name":"Yan Tseitlin","email":"yan@lightbitslabs.com","username":"yants"},"state":"CC"},{"updated":"2025-12-11 17:45:47.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-12-12 22:06:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-12-13 21:08:48.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"26bb73da3da4a87e00b3f3cdbb85161dfd82e0ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 17:40:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6760378099cb18137f1783e67087b15f62cfdcd3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-31 17:50:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"70f10a5cb759d50f36b743615d3b3c17bec0ea42","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-31 19:24:21.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0d4ed51932ba94abb59e650fa726c877b94656b1 .","accounts_in_message":[],"_revision_number":1},{"id":"806b84bd263106706c409f6b83ca5d47463ff5ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 20:10:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2441f4bb78f04fdbae6275bfadb52f1d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cf3151bc13a1476cb66c2c82ff09b6e9 : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e734e79c04de4275ae8b818f42cd9823 : NODE_FAILURE Node(set) request b7fd0a8243cd4f2aabe5077a695535e9 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0084c8ded50e426e934413b0b07cd283 : SUCCESS in 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf48344c2022e93d4a585524757072dd68b89e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 20:43:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6cdfed59265419e8d3242f2781c3fe3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9b803bb6b4a4c3a8fd4e2c8b8f88b74 : FAILURE in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/725152fa3044422baf25713efeb35658 : SUCCESS in 9m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e30f7b5c2b8945c9bc9538b7b9becbfa : SUCCESS in 15m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b9e85ca3d6d4f0e80bddc8b96862a8c : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/2ce196d2d306429d837995e07f38c7e5 : SUCCESS in 57m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5a4f18568c9242d190aed900c426459e : SUCCESS in 1h 00m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/648c36bfa070422ea2f093308f0369a6 : SUCCESS in 1h 45m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/618a16c3bf8f43f0b6136289e11c68dd : SUCCESS in 54m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a103d39c61a94801b209ac753fff337d : TIMED_OUT in 41m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ab35391a794c4a0590e59b1c51fba22f : SUCCESS in 5m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1cb14de7e4894367b00b3f735a299a3a : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7d43def1b0e448969b222d91be708b64 : FAILURE in 6m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/21cd68f30c484568b8d2a41d0acc4aed : FAILURE in 9m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/37a5ac7febb741aca102567c0b254403 : FAILURE in 18m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db1ca6c1ba4d45dc89f86d29e8b1a5ea : SUCCESS in 6m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/65846284678648aeae9b3ede1ddca310 : SUCCESS in 1h 33m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b56e7495d39e410b81308c039c9396b2 : FAILURE in 2h 55m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/802882d0d50c48d6a4cd03a9961f691a : SUCCESS in 1h 40m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/79c3096a9701440ca6877623998cb931 : SUCCESS in 1h 36m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d5ce8de5c3a443cebca9812af0507932 : SUCCESS in 18m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/16869ddd9ad941f38ac5a2e59b5655c9 : SUCCESS in 1h 07m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e04f53ec879b48b7bb9272e1e8ee802a : SUCCESS in 2h 21m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9a5c6b3e475b4e45b8e385d022ed8bae : SUCCESS in 35m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/14148a397d394b4bbc8344041904a6b4 : SUCCESS in 2h 00m 13s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/67396ab698204b3eab8a79e27d15a8f8 : SUCCESS in 2h 32m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d6f8926b60fa4e2a8645995e657d2575 : SUCCESS in 1h 48m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba9dbec5ff8147479b9a622c9104e848 : SUCCESS in 1h 11m 56s\n\nWarning:\n  Comments left for invalid file cinder/backup/rpcapi.py","accounts_in_message":[],"_revision_number":1},{"id":"a4fdf479a9bec03cd48a296ca09fa711622a443a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-01 16:21:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2271/.","accounts_in_message":[],"_revision_number":1},{"id":"9527c6dedbfa854d2c40d592c1ad1dd780981688","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-01 16:34:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/965841 : FAILURE in 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"68809db99b9c0e3d27d043c83e09f7587ed2b78f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-01 18:06:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2385/.","accounts_in_message":[],"_revision_number":1},{"id":"627420ca22b01b140814440456a2db0a2dd2a7d8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-02 08:34:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0d23416533e7470ab77a45de5a6e658a : SUCCESS in 1h 41m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0e3ca8d52aa74700a47ea0e9f2800ec0 : FAILURE in 1h 47m 14s","accounts_in_message":[],"_revision_number":1},{"id":"19d1edb60337a9f5149840784cf433baf44bec8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-03 17:28:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7a4bb7de75490edd4eea835a30a748b5ea5397c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-03 17:34:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"83480726cf31be2c493f600914a0fc410053275f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 17:49:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36da224b42a14f9d98ab2e01cc70939a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/334a891670704ade861eda9cb2859b95 : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e3bedd2417b4c96810acd3559e128b0 : SUCCESS in 14m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/77c61c2dc76647cca0bfa8ef7f28cdbd : SUCCESS in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0d456a5d38872e62389c8bebe088ceab9058c0fd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-03 18:27:52.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0d4ed51932ba94abb59e650fa726c877b94656b1 .","accounts_in_message":[],"_revision_number":2},{"id":"b31d8ff95e539932388ea97400243794380a3553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 21:09:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e569619bf254d89b1b7afffe9acbcd0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d62c37725cc49088ab1e096318c2b25 : SUCCESS in 7m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f900d9a29f440739c62a5318684f119 : SUCCESS in 13m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e95478133a6c469eb9eb3795b1dda441 : SUCCESS in 17m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99fbd77ba14a4853b31e0b6236734657 : SUCCESS in 15m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/49ef99fc00fb4ffd8cfe4f41143dc658 : SUCCESS in 1h 01m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/746d04118a154ec8828339618a9602b6 : SUCCESS in 48m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b6a2a59acbfd4b26a5fc781642264478 : SUCCESS in 1h 36m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94570730b581441aaecdabdb3eb0ef5d : SUCCESS in 58m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f28b738b296644209ab7a263895613b9 : SUCCESS in 23m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7c0acb54389245a59f5667486ce9a94f : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/827d096b55c04386b40af9387ee28e8e : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/083daa692e944d6fa9b6e7991133b3f8 : SUCCESS in 5m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/df417aae4af2420c975ff8341c6712b4 : SUCCESS in 9m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7de792c09e24485fa6b353f3c1f90349 : FAILURE in 42m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/377f30cd74ab45a7a98d277e5ae37ca1 : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2d4a8a95c8bf4faab8f9a76fb55d83dd : SUCCESS in 1h 52m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d0cd947287c84eaa83320107b93c37db : FAILURE in 2h 43m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/81c2ea0278a94d66862739934dafc347 : SUCCESS in 2h 08m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c606c4e9e5e5405287c95e06b51618cf : FAILURE in 1h 58m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/815fe25913cf433e9d631e1c7cdca077 : SUCCESS in 37m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/412a0844e6d74999ad790a663e4bb7e1 : SUCCESS in 1h 10m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a33bcfc2bba249c888ea8cea13bd7630 : SUCCESS in 1h 51m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a3bb9d72c92b47e7b4d610a7afd9b3fe : SUCCESS in 43m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d444dde889d049dcb53351fbbadb5ebb : SUCCESS in 1h 45m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/c7933f2010e24120b53ff6c02cd9704f : TIMED_OUT in 3h 34m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/96e3457480394811b7a3a8b3820f5ec1 : SUCCESS in 1h 17m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47d84932f70a4bdb88ca080fdca4f84d : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":2},{"id":"65793a44dd243646a5c1e029914084a08f06f87e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-11-03 21:15:59.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_11_2025_21_09_32\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"28d4f635e6711f0446ac8307f9af35f9e382bcfb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-04 01:30:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/965841 : FAILURE in 36m 35s","accounts_in_message":[],"_revision_number":2},{"id":"806fbfacbfdbf84428c92f9143f3b9a937935e7b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-11-24 16:55:57.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_11_2025_16_49_29\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"79521062e6dde0eca940aebadeb44d44df4acbc9","author":{"_account_id":34470,"name":"Yan Tseitlin","email":"yan@lightbitslabs.com","username":"yants"},"date":"2025-12-02 13:14:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5d1371cab4a9ad9ffba226b918b885e92cb1e09c","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-12-11 17:45:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3ff3843a7f88a80df3e43a758ab6c6fb4815bbc1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-12 22:06:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ae42843234ca2b1e827c644f2231ac0c1da5780","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 22:07:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"62f8163b471a7170d50db258c05b702d98926974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 23:47:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/956948388abb4d47a4bf5dfa67a69765\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2cbe3317d354a5ba84b8a1b6f42adce : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/864c72d1709d4589bf9dc5a03ec04617 : SUCCESS in 11m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d630e5b77ae4a0590b2fb6aaf6b389c : SUCCESS in 14m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b4d1a23969c4dd2910bae89bfacde4b : SUCCESS in 17m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/389e906c4a8747b8ba6d58f61dcc30ef : SUCCESS in 1h 05m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3137337a3af04317b9a6c0b72c78a977 : SUCCESS in 48m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c435f0545c764d66a77072b458d41e4d : SUCCESS in 1h 35m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdc809ae80a04ba4b050e65012b8dbcb : SUCCESS in 1h 01m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d709b742900d4672b73a11789efbff06 : SUCCESS in 59m 47s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dcdb3a88454d4474930a8f1a7391cd2a : SUCCESS in 59m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06691985a55f441ab0e35bad67fb2a98 : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ba73d4bc2c103ea41238ef8c61cd9617f03a84e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 23:47:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"205dd985ac5dc1b5d04b073c5a5dee452dc82af1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 23:48:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f85ab3deda847d4bab8c86f5be92fae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d08da68014b4a0fbd9d2e49ce83a6e8 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2},{"id":"bea4235fd1383fd4c712ef6be1f36a6250538900","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-13 21:08:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-41-965841-2-fc : SUCCESS in 52m 8s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-41-965841-2-iscsi : SUCCESS in 1h 4m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1d41db9f89e8d394f47699d24f0232c3bf768baa","revisions":{"86c65afd1191861e82fe279c08c600994d3590f4":{"kind":"REWORK","_number":1,"created":"2025-10-31 17:40:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/41/965841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/41/965841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/965841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/965841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/965841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/41/965841/1"}}},"commit":{"parents":[{"commit":"c34bd8ef77a6e7baee6eb6cd5cc7108e08d7e115","subject":"api: Rename controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c34bd8ef77a6e7baee6eb6cd5cc7108e08d7e115"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:33:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-31 17:39:40.000000000","tz":0},"subject":"api: Add remaining missing request body schemas","message":"api: Add remaining missing request body schemas\n\nThis is relatively simple. The schemas added are either (a) extremely\nloose or (b) merely mimic validation that already exists in the API\nmethods. Where the latter occurs, we remove the existing validation\nlogic, which effectively means removing invocations of\n\u0027assert_valid_body\u0027 and \u0027validate_name_and_description\u0027. So far, so\ngood. We also rename the imports for consistency as we plan to conduct\nfar more surgery on these.\n\nWe do, however, uncover a bug. The \u0027VolumeTypesManageController._delete\u0027\nmethod is getting mapped to two APIs: \u0027DELETE /types/{id}\u0027 and \u0027POST\n/types/{id}/action\u0027. Only the former is expected and attempting to use\nthe latter fails, but it fails with HTTP 500 (bad!) rather than HTTP 404\n(good!). This can be reproduced with e.g.\n\n  \u003e\u003e\u003e import openstack\n  \u003e\u003e\u003e conn \u003d openstack.connect()\n  \u003e\u003e\u003e type_id \u003d \u0027xxx\u0027  # a valid type ID here\n  \u003e\u003e\u003e body \u003d {\u0027delete\u0027: None}\n  \u003e\u003e\u003e conn.block_storage.post(f\u0027/types/{type_id}/action\u0027, json\u003dbody)\n\nWe should fix this, but doing so is out-of-scope for this change so a\nFIXME is included for now instead.\n\nOn top of all the above, the schema decorator we have expects a \u0027body\u0027\nkwarg to be passed to the method and requires a request object with API\nversion attributes. We had a couple of tests that were passing body as a\npositional argument instead or using fake request objects. We correct\nall these issues.\n\nChange-Id: I0d4ed51932ba94abb59e650fa726c877b94656b1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/86c65afd1191861e82fe279c08c600994d3590f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/86c65afd1191861e82fe279c08c600994d3590f4"}]},"branch":"refs/heads/master"},"1d41db9f89e8d394f47699d24f0232c3bf768baa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-03 17:28:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/41/965841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/41/965841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/965841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/965841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/965841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/41/965841/2"}}},"commit":{"parents":[{"commit":"75d12bc023e31250a1cabb87e6c1b69e03c7012a","subject":"api: Rename controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75d12bc023e31250a1cabb87e6c1b69e03c7012a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-30 12:33:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-03 16:07:24.000000000","tz":0},"subject":"api: Add remaining missing request body schemas","message":"api: Add remaining missing request body schemas\n\nThis is relatively simple. The schemas added are either (a) extremely\nloose or (b) merely mimic validation that already exists in the API\nmethods. Where the latter occurs, we remove the existing validation\nlogic, which effectively means removing invocations of\n\u0027assert_valid_body\u0027 and \u0027validate_name_and_description\u0027. So far, so\ngood. We also rename the imports for consistency as we plan to conduct\nfar more surgery on these.\n\nWe do, however, uncover a bug. The \u0027VolumeTypesManageController._delete\u0027\nmethod is getting mapped to two APIs: \u0027DELETE /types/{id}\u0027 and \u0027POST\n/types/{id}/action\u0027. Only the former is expected and attempting to use\nthe latter fails, but it fails with HTTP 500 (bad!) rather than HTTP 404\n(good!). This can be reproduced with e.g.\n\n  \u003e\u003e\u003e import openstack\n  \u003e\u003e\u003e conn \u003d openstack.connect()\n  \u003e\u003e\u003e type_id \u003d \u0027xxx\u0027  # a valid type ID here\n  \u003e\u003e\u003e body \u003d {\u0027delete\u0027: None}\n  \u003e\u003e\u003e conn.block_storage.post(f\u0027/types/{type_id}/action\u0027, json\u003dbody)\n\nWe should fix this, but doing so is out-of-scope for this change so a\nFIXME is included for now instead.\n\nOn top of all the above, the schema decorator we have expects a \u0027body\u0027\nkwarg to be passed to the method and requires a request object with API\nversion attributes. We had a couple of tests that were passing body as a\npositional argument instead or using fake request objects. We correct\nall these issues.\n\nChange-Id: I0d4ed51932ba94abb59e650fa726c877b94656b1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d41db9f89e8d394f47699d24f0232c3bf768baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d41db9f89e8d394f47699d24f0232c3bf768baa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
