)]}'
{"id":"openstack%2Fcinder~965842","triplet_id":"openstack%2Fcinder~master~Ie09de58bd1b01edea916d7e43e32eb81e7877fae","project":"openstack/cinder","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-12-12 23:59:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie09de58bd1b01edea916d7e43e32eb81e7877fae","subject":"api: Remove unnecessary validation helpers","status":"MERGED","created":"2025-10-31 17:40:18.000000000","updated":"2025-12-13 09:40:19.000000000","submitted":"2025-12-12 23:59:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965842-openapi","meta_rev_id":"b1ece91cb5a697ff058f637db50e03161cc781cc","_number":965842,"virtual_id_number":965842,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-12 23:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-12-12 22:07:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-12-11 17:46:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-12-12 22:07:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-31 17:52:06.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 19:36:40.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 19:57:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-01 17:51:04.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-01 18:21:12.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-02 10:26:03.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-03 20:04:51.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-12-11 17:46:38.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-12-12 22:07:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-12-12 22:07:32.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2025-12-13 09:40:19.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"6a06a297642f0b504ef3fc7aae9f90e87281ffc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 17:40:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5caf2ea30c49e0ff09df4f859087cb0c552cc458","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-31 17:52:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"abe1437da6edf1f4bd83073745a6a6e09953f9fc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-31 19:36:40.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie09de58bd1b01edea916d7e43e32eb81e7877fae .","accounts_in_message":[],"_revision_number":1},{"id":"7712c6229ceee0ee76440070c775c6218623c256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 19:57:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e732c2b361b49f487a9d8ad2bcba7a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f6ecd3de6c4a0b977f1cbd5b9ee404 : FAILURE in 8m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77ec8341d5034df0b758c534172a333c : SUCCESS in 8m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e04e78db213484fabbca0cc8c56671e : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/985edfaf9ebd440fa769c62d8c079941 : SUCCESS in 14m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/f6209427b8084a5e844df020bc47499f : SUCCESS in 58m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c446920e107409c8cf9300257032204 : SUCCESS in 1h 12m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/74f72e33cda541edbd6e324685d66cbd : SUCCESS in 1h 38m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bb206929f81412ba0057c51605d2b78 : SUCCESS in 33m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/296590215325486bbb19fab76a4a2cd7 : TIMED_OUT in 41m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/19f8f4d0b6ac45eea23b3e4079510fda : SUCCESS in 5m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e974853a748b4f8f931afc49ac536e6f : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d72cffd6bcf04301b56095fa6bd27d75 : FAILURE in 6m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1a52a66af59740feb6b68c84f4f0ddc9 : FAILURE in 6m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1bd0123bc4ac4da18101a41e167804bd : FAILURE in 41m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/36638da647824a2093b90b6e86eeaa20 : SUCCESS in 8m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/692df17564824c2bbc862f8990850430 : SUCCESS in 1h 45m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/86230a4d3833455ab6986ae19e6118cd : FAILURE in 44m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3a733b349ee8451a87ea9335a9ec68d7 : SUCCESS in 1h 48m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b983e9e33f214bab91c87cc3daf50be8 : SUCCESS in 2h 09m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/98473fbaa4a340d2b121ad8766038c08 : SUCCESS in 29m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/af43f0d0e39448f186a0eb9c64627a8f : SUCCESS in 41m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/04768946b5ec47b082fa3f113731742f : SUCCESS in 1h 31m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bc8280ef8bd547d6b19d0d85053dfed1 : SUCCESS in 47m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/68d67e62c54c41e18362b92b3e30f018 : SUCCESS in 2h 09m 11s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/7d4a13efc56742b482605989156a4a58 : SUCCESS in 1h 50m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/44fa3851bf0c409f809e1f68d4da5932 : SUCCESS in 52m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/328a8b22ae424cdb986ad28ae3b24143 : SUCCESS in 57m 55s\n\nWarning:\n  Comments left for invalid file cinder/backup/rpcapi.py","accounts_in_message":[],"_revision_number":1},{"id":"524b96aad1d0c3a6c714b13e06eb937414b67155","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 20:12:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2b68b38cf39458dbae4713884a98185\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/631374ca53ff45abbbb17640c62b364f : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3fd4f5f4f1cd4daa85054d9503d23699 : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b46648522f7b4d0f9e4ed7ff59166677 : SUCCESS in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7185ae205894cf3a11b32e37d82ad66c156ab802","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-01 17:51:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/965842 : FAILURE in 1h 11m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1d108d50445958f546aedf610eb38880c28c81f5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-01 18:21:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2273/.","accounts_in_message":[],"_revision_number":1},{"id":"ea40c2762b9f0c18b448079fac84283cfb866f23","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-01 19:31:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2387/.","accounts_in_message":[],"_revision_number":1},{"id":"e065c13177c11fbc36f8f1c8afa443b75fbb1ab7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-02 10:26:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cf104a42038043da9cc6a609513eb061 : SUCCESS in 1h 42m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cd197240bdfb4bb28ceb477cb1552d1e : FAILURE in 1h 50m 55s","accounts_in_message":[],"_revision_number":1},{"id":"03e8b66eca53624905cf8c267c42497776e8ef82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-03 17:28:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f2625aa8f4eeba7a1c98cb886eb1e2801d0ee735","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-03 17:36:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"fa7d611992d92f5d5e2f143c82a64bce04db1415","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 18:19:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f144fe5baa245278e24ea96ff77a9c4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d1e06265834048dab5fd5d8bb224e71f : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9dd44c0754074c738e7cb4043b80fe6d : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b0504bd0aa764223aca180940c286a87 : SUCCESS in 23m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34654feddba190714fa1cde797712d9e4a66487b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-03 18:38:55.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ie09de58bd1b01edea916d7e43e32eb81e7877fae .","accounts_in_message":[],"_revision_number":2},{"id":"e00499f187f871898af9fdc916d6ab86697d404c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 19:58:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf652a8b47ef48698ce915c2b9193dc9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd78deda7494cb3860ea0a29fc79b7f : SUCCESS in 7m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/657106c12c1e46a0b0c532ed9f31044f : SUCCESS in 11m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cba93d5188c84b67bbf3bbab47a71f5f : SUCCESS in 15m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f298ae571ac40789cc65a7a689d0ef1 : SUCCESS in 14m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/7f68bfa0daf942eebd49c37fbaf53f4e : SUCCESS in 52m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f269ed03569641afacdc352cc677578e : SUCCESS in 55m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/db4e1b2b026b423cb72b7e00146be19b : SUCCESS in 1h 33m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51e49b45b9f140518f0b72215f6121a0 : SUCCESS in 1h 02m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1501000a118c4294ba0ebbb43cd09f36 : FAILURE in 24m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/627bee6dfdc2404191de04847e9f8122 : SUCCESS in 7m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3b608fe56a6d49e181237f94024b39cf : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bcb529cb6421489bbde6311d00cc3201 : SUCCESS in 6m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a872ca31d9df459ba267638e4c0ed445 : SUCCESS in 9m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bf0ba98f80bd46808bff25195a145d7b : FAILURE in 42m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c27b8e80f8f24377a375504ed5e98fdc : SUCCESS in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/43e49488ed004ddeae73e8f1e0ead773 : SUCCESS in 1h 47m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c3c1d24473c64631867c5c1c459dd816 : FAILURE in 1h 42m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b383e3b55d2447cc8fa74be3194e6ef1 : SUCCESS in 1h 58m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8b92358bf7b84a789de0cf22474e86fb : SUCCESS in 2h 13m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0a42c0dd72dd45b592c7768af337cf6e : SUCCESS in 31m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e247603a26df4249b7cbd8a1a9148a03 : SUCCESS in 1h 13m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d5ccda6de219451999d0d036cb1f505c : SUCCESS in 1h 29m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1d676ed7e4b94c629dfdcc0f9dbf8eba : SUCCESS in 44m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3df4b5dbb39e4e1eb291e91748e822d0 : SUCCESS in 2h 02m 43s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d3582ead563640e6a850d2f4da4d0515 : SUCCESS in 1h 43m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5ba3a1d2022d4f7db73381eb540927c0 : SUCCESS in 1h 44m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f32e55632774a51ac201d6efa8779e5 : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e76b7a0aac23ddc4726aeba3b9d17ab3fb293569","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-11-03 20:04:51.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_11_2025_19_58_07\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"33cb9494f25dea706e1288a3e71d15d8c36225a5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-04 02:48:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/965842 : FAILURE in 1h 13m 08s","accounts_in_message":[],"_revision_number":2},{"id":"fead07b438fff6859496f5a3b844044ad162395b","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-12-11 17:46:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"82f3fecd4f88111bbeea013c4fccc02c6f7a5149","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-12 22:07:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b82f59b302dcdb4435998b17e7e4e3c71439ea88","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-12-12 22:07:32.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/965841 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"7701a00887254d01b8f48d936bb3573045120dc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 22:07:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a532ad618ed82c5c87ce0d13f2d49318237c248","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2025-12-12 22:07:48.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/965841 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"fc04fc73bd5d7cac616bba83c6bccd2cfc85da63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 23:59:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de7d8c19a2584298866002f395c77f34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b4006e6a091431aba12e3dfce4e5154 : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/866645b6da3b430185f5743aa81a6248 : SUCCESS in 7m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b6f6e8e21644b718dcf9a9b4d367622 : SUCCESS in 15m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c5eac861c8943a5bba9f6af118a0dc8 : SUCCESS in 7m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/def9a70d4d1a4ae2bb84f5570a4423e9 : SUCCESS in 31m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bfa115432f8d4aeaa4cd87e062e11cd7 : SUCCESS in 1h 00m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ffeed89a29449cb833ee52909197d42 : SUCCESS in 1h 37m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f0be9a56e714d49bf51266e57ee36e7 : SUCCESS in 1h 04m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/79e761d55f774685839241a3720dfdd6 : SUCCESS in 35m 19s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e5c788be15eb4da6b199b3691d6af132 : SUCCESS in 1h 46m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a030621f93a748c688d0e4e704ba2ba6 : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a02dcc892a371a3e945f4bc8dc0d8d4c291e45bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 23:59:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ffbdc62e0e2c3ebe9a81541ca57206b97d36ef0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 00:00:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2059f0eda634b3db1b0365676fbb8b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e33cabb05048939766f861cc2030cd : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2},{"id":"b1ece91cb5a697ff058f637db50e03161cc781cc","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-13 09:40:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-42-965842-2-fc : SUCCESS in 1h 18m 58s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-42-965842-2-iscsi : SUCCESS in 1h 8m 3s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd04e532fcb4de1adce1aa15fcd30d6da4e2e01a","revisions":{"de26b79f7e5b00fc43cd801093ca075c6b7fc5fe":{"kind":"REWORK","_number":1,"created":"2025-10-31 17:40:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/965842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/965842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/965842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/965842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/965842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/965842/1"}}},"commit":{"parents":[{"commit":"86c65afd1191861e82fe279c08c600994d3590f4","subject":"api: Add remaining missing request body schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/86c65afd1191861e82fe279c08c600994d3590f4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-24 11:20:30.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-31 17:39:58.000000000","tz":0},"subject":"api: Remove unnecessary validation helpers","message":"api: Remove unnecessary validation helpers\n\nNow that we have JSON schemas for all APIs that accept a body, we no\nlonger need to do this validation in the method.\n\nChange-Id: Ie09de58bd1b01edea916d7e43e32eb81e7877fae\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de26b79f7e5b00fc43cd801093ca075c6b7fc5fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de26b79f7e5b00fc43cd801093ca075c6b7fc5fe"}]},"branch":"refs/heads/master"},"bd04e532fcb4de1adce1aa15fcd30d6da4e2e01a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-03 17:28:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/965842/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/965842/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/965842/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/965842/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/965842/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/965842/2"}}},"commit":{"parents":[{"commit":"1d41db9f89e8d394f47699d24f0232c3bf768baa","subject":"api: Add remaining missing request body schemas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d41db9f89e8d394f47699d24f0232c3bf768baa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-24 11:20:30.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-03 16:07:24.000000000","tz":0},"subject":"api: Remove unnecessary validation helpers","message":"api: Remove unnecessary validation helpers\n\nNow that we have JSON schemas for all APIs that accept a body, we no\nlonger need to do this validation in the method.\n\nChange-Id: Ie09de58bd1b01edea916d7e43e32eb81e7877fae\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd04e532fcb4de1adce1aa15fcd30d6da4e2e01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd04e532fcb4de1adce1aa15fcd30d6da4e2e01a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
