)]}'
{"id":"openstack%2Fcinder~966186","triplet_id":"openstack%2Fcinder~master~I7398adf8e35709cebe7e1348b2bf2378f23f893e","project":"openstack/cinder","branch":"master","topic":"bug_1870367_qb_3","attention_set":{"13915":{"account":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"last_update":"2026-01-14 13:46:09.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7398adf8e35709cebe7e1348b2bf2378f23f893e","subject":"quobyte: Fix extending attached volumes #1870367","status":"NEW","created":"2025-11-05 14:45:19.000000000","updated":"2026-01-14 20:18:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"298beab9242a8cdee634c9e3f12a25806cb45c34","_number":966186,"virtual_id_number":966186,"owner":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-14 13:45:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-05 14:46:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-05 15:41:46.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-05 15:50:14.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-05 17:24:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-05 18:03:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-05 20:18:19.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-06 06:08:18.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-01-13 16:40:25.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-13 18:06:48.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-13 18:06:59.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"}],"messages":[{"id":"32a4b484c7b39037cd1f13d9adb7ba2525386a40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-05 14:45:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae772908c0c43d86469755db7c6c042a8697b206","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-05 14:46:23.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"85a5bf66c94694ad4894ac9b2f2f40941c10302a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-05 14:46:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"982af3feb43f90c5f068c89a11e6ac1e9394e67f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-05 14:53:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0edf80969e59313ef53af132f4f2cd17b70c4004","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-05 14:55:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"368f3168af91954ee88dcee56e62f80943df58b8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-05 15:41:46.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7398adf8e35709cebe7e1348b2bf2378f23f893e .","accounts_in_message":[],"_revision_number":1},{"id":"c643469267a43f07cc8eba193a8b3a97ae908e65","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-05 15:50:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d58674c11a5746cba2187b1cd77eab85 : SUCCESS in 47m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0a0df66a99a2490b9a3c34ee80342717 : FAILURE in 52m 34s","accounts_in_message":[],"_revision_number":2},{"id":"7ae8a2b174cd94e39daf17e9a87a9f7b16c8b5e8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-05 15:58:26.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7398adf8e35709cebe7e1348b2bf2378f23f893e .","accounts_in_message":[],"_revision_number":2},{"id":"4c1b8dbf12d8acfb603d20396f9cf40892bd4a39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 17:24:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aac294b4bc984ec7acc9906a54199428\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c602e15213e43fe871947c650496f82 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/536b1c95aaf24ea89bb180800f2b98db : SUCCESS in 25m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/76508bfc61bc4461acc22e28df4461cb : SUCCESS in 26m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a43251cdb60ad81d4ec5a0f7f478ac103b489234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 18:03:54.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b20d7f8e1cc441e2a4b61ca3abff9dbd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42edbedbd42b47438b214c61c66fa3f2 : FAILURE in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09ec095e691c4fe4a76fe1e6281e3c03 : SUCCESS in 11m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80fecbf0a3d94a58a56aa3b79497f503 : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e5990ad1534580b10a1feb939e6a46 : SUCCESS in 14m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/bdeb094259e94a219fb11142f4e08f97 : SUCCESS in 1h 01m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a936c31e4dde4fc2be1c1e0294c99d32 : SUCCESS in 1h 01m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/423061e458644b0bb6b66123b18bd603 : SUCCESS in 1h 34m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8ed8aa7df564932bb8fe358cfc7584a : SUCCESS in 56m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/86ab7702affb462d87c2d48b18257f56 : TIMED_OUT in 41m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/239b8644d3a648c8a9bf096a69c85671 : SUCCESS in 6m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e548c8dc2c5940b1a8df93959ce16e48 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/67f8409f1201418cbdbb8a969e652c1a : SUCCESS in 6m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f181b05b00d646068e75a581dc5977ee : SUCCESS in 8m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/15ced6be43594d39bc5a446a841d0a40 : FAILURE in 41m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b0683e3707a04d02b488d8c6f741d3c8 : SUCCESS in 7m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b2c2155cf4304332a0f7457f2503ff4c : SUCCESS in 2h 02m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1bdfdd57a1d045078984fb3d5c755d9f : FAILURE in 3h 01m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ef02f2081b7a49b9bf86250222fab8da : SUCCESS in 2h 24m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f9a6abafadfe42b69e2feb1bed5680ff : SUCCESS in 2h 14m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/653f32d20d774b6eb69577e95b7f261f : SUCCESS in 38m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/57abc80ed8b94012a40f6c39bc141495 : SUCCESS in 1h 10m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8b4f395e92524a0fb89c1a38ad6b5e18 : SUCCESS in 1h 42m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f393f2d910da46ee93c5f6b730741f00 : SUCCESS in 40m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/709afc8f5e5b4ce5828413127343ca68 : SUCCESS in 2h 01m 19s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ec99730f3cd3440bb19ca6c1d56c8b58 : SUCCESS in 1h 59m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6076ee031fb041f9bacac94aa1d7e268 : SUCCESS in 2h 01m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56efb5f6820e403ba4fe6f0b637a84b2 : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":2},{"id":"cf0ecbb35b7efa4ef7b58de17c29ce712648077c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-05 20:18:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966186 : FAILURE in 1h 13m 43s","accounts_in_message":[],"_revision_number":2},{"id":"5098acc1fc796b09da6262a1a2e226f85473edda","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-06 06:08:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2431/.","accounts_in_message":[],"_revision_number":1},{"id":"ea8205a1ca0461da88bcd49b6fb20ae3a0081dc9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-06 06:10:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2317/.","accounts_in_message":[],"_revision_number":1},{"id":"21e74f31f81d9cc3d4cab7dcc584515805e8155c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-06 06:27:00.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2318/.","accounts_in_message":[],"_revision_number":2},{"id":"d0471e76efc1a64974201504bc103ba1f2a29065","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-06 07:10:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2432/.","accounts_in_message":[],"_revision_number":2},{"id":"d3d97606706e6638639ab161ec1483eb9b7a718b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-06 13:50:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b176e3f45a4c31aaabc185b41ceb309b07e83764","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-06 13:51:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"d1d0639dc8e000300f31cf8fee824c09e86d0a47","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-06 14:47:15.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7398adf8e35709cebe7e1348b2bf2378f23f893e .","accounts_in_message":[],"_revision_number":3},{"id":"e662a3ba746f2bae332a84513e70f9aa17b16405","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-06 14:47:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6a06c5331809431dac34de9e06a8a6ef : SUCCESS in 52m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/88306d1ee8b54b89afd71fdf372d295d : FAILURE in 52m 55s","accounts_in_message":[],"_revision_number":3},{"id":"4e8f0c1921882041db9d1ddd0457bf4c1b061699","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-06 15:05:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966186 : FAILURE in 1h 11m 50s","accounts_in_message":[],"_revision_number":3},{"id":"b6494228d29b1807b2b3815f761acfe640ca6182","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-06 15:26:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e4a2d5a9c73e650b35955e5771a84a8d3c4934b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 15:41:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b05bd1ba5eb456b9191f8cbf97b79b8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fa7c6a274c0648f18124a2ee73b8a666 : SUCCESS in 14m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c0a17ab4d2f54c21bb57d2edb7f0cb6a : SUCCESS in 26m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/db32dbf1991045c297e83f3737a44ac4 : SUCCESS in 22m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a58715313a564bff5244c7a782310c9a9f4d92d4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-06 15:50:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"8d484264200fc5173cdfcc394bf0dd81f6dfb165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 16:14:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68d6d354312d4cb3b35c8ecd200f726f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49fc84eadb924a3096e740e7ea5eae46 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a835375f5eae4c6aadc54ee56d39480b : SUCCESS in 11m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bd48264e4d9430db3894851132a18fa : TIMED_OUT in 41m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab5d9f5b7aa440f8d420f7401156476 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/fede852e68a04f7fb2342d3e301e0667 : SUCCESS in 57m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00bcb74f182242f4952f2fb3f7411596 : SUCCESS in 43m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1be4b8d60a11473aa138d466c871ada9 : SUCCESS in 2h 04m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65a86a567a574124be7edd3e392eaba8 : SUCCESS in 39m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fc0c6096c3344ac7838a1729db584f73 : SUCCESS in 20m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8ed7c6515e2c4e7980b85b0ef5731284 : SUCCESS in 6m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9e3fe21ba0c448bb924614ece1cc80d : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a78530b861704f9baf64ac7fc4558556 : SUCCESS in 5m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aadcf4be0ed84834b281f3f5472d7b03 : SUCCESS in 8m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0759452c4fce48218daf2d6db50c7ad9 : FAILURE in 43m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f6f55c210b14751bbbe1e24004a189d : SUCCESS in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/521063cfeea544cfbc111d99df9ba9c6 : SUCCESS in 2h 09m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/57024b152cae4b17a7dafc07b32fdad6 : FAILURE in 1h 37m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d57e050d77bb4aa682a555166921dcf0 : SUCCESS in 1h 43m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/af9938ebac3242999d39596240a3c7b3 : SUCCESS in 2h 22m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/472444cc8e9d481bbbe2c94706f95d47 : SUCCESS in 23m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d4a47ab5a5684f6abf047ebcf6af418c : SUCCESS in 1h 09m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a23f44da357f43a39e4d231d49c73382 : SUCCESS in 1h 52m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/abfd59ef67404eb0a844a06ffd7be4a4 : SUCCESS in 35m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d8882b7925084b0b844fbb4ed07818fc : SUCCESS in 2h 19m 15s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/df89a093b6004f42b103b1776cec40ac : FAILURE in 1h 52m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9bba24aa2a44b16af292603ad39bc10 : SUCCESS in 1h 36m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d558c3777faf45edb4f3ea8bfeaa082f : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":3},{"id":"07e95137c7f53f0e798a27e318aa7afb018bae26","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-06 16:31:17.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7398adf8e35709cebe7e1348b2bf2378f23f893e .","accounts_in_message":[],"_revision_number":3},{"id":"16ecbcbfbc8c051c3070c770db6e3bb543c82ecf","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-06 16:56:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"195d1d9bb31708f72251950fd7b4e32719ebc20b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-06 17:34:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2330/.","accounts_in_message":[],"_revision_number":3},{"id":"899bde0465a10ffa036fe18dcb48c3747ba42dc4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-06 17:35:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2444/.","accounts_in_message":[],"_revision_number":3},{"id":"a0819a2ff2c6ac942d340b07b0bc7c850cf9b02f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 17:37:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2b95ed245f44767a51508c8d6b64e48\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/79b32f710e6044b6bbb818ddf97a5023 : SUCCESS in 22m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c25e64be81e64bdf9b9e405742284b8e : SUCCESS in 24m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c8d4d8952964eaab84219ca0a8bacf7 : SUCCESS in 22m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"226d485dc992a4dc6054318a50d8c1b34103e849","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-06 19:02:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bf42c65e9faa4a73adf5857ad795e4fe : SUCCESS in 1h 41m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/60b100db093244298548e84aafd0e5a9 : FAILURE in 2h 02m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8db88af2857b8d5489306d4d05e3da74a339812c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 19:43:10.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3df4dcbc8d844f681d200e2a33e78e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87f2876337894648acca1ac57c8874ec : SUCCESS in 7m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82aed8fb7ced4db0ae11f65471ed0752 : SUCCESS in 11m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ac97dc608494d6295587c362eaa84a7 : TIMED_OUT in 41m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cca02fdc25ce4f38805d412067c728de : SUCCESS in 13m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/038f724d583949d8911fce3e57058762 : SUCCESS in 55m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cfbd4fc3f7e445aca82d4a71a153f90d : SUCCESS in 59m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4f883bc992c48518dd53907a33c08e0 : SUCCESS in 1h 26m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87037b2a5c254fc281cf4a1962c0f71e : SUCCESS in 58m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/29f93c76914d4b1c88a5b2c077d0f553 : SUCCESS in 22m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3ab011f9f8f543a59031df4a8f58e1b6 : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f64d0ab7713d45b1bed58d77b4e5446d : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/845295b37cd24e6c98f60815ee123d4b : SUCCESS in 6m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/617daef09cb84a4fa0e1ef653a9f2337 : SUCCESS in 9m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b53d82f512654e83a6c3b2c027b52ead : FAILURE in 24m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e32f2b2606c545bd9e5b4a99c7c81e24 : SUCCESS in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bc7e7f0271e047a09bf3e61a81a28ebf : SUCCESS in 1h 53m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fc7465e6611046d8bf7530428447acef : FAILURE in 2h 41m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/20253b082ff6485089555b1dee41c307 : SUCCESS in 2h 00m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e82735742a3847a7bb1e704a3660c8e1 : SUCCESS in 1h 53m 47s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5ce6c66a16dc46f4ad434b3eaa1f4432 : SUCCESS in 41m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5b5f7a133991486d9aef3977612f61fc : SUCCESS in 1h 13m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/58ae0d99e5fe437695a54d9594c7361f : SUCCESS in 1h 49m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5475f43f560c4232bfbe05b4f5c9ed08 : SUCCESS in 33m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e7eb64e0989c405fbc5b12730cd4d3f6 : SUCCESS in 1h 18m 20s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/479bfc4109ac48a99d97306ad735c6b5 : SUCCESS in 2h 31m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e2c2a4870b004c2e8486c9faca91d350 : SUCCESS in 58m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6e2f4d3e1d84185b31fb6bd05d5b81d : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":3},{"id":"f1d82f3ea1562538c05beaffd5977c4cdaede7e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-07 13:16:12.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a3b114c02a6eb80013ed4a2bccb2867f5903d60b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-07 13:17:49.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"a8239156ab91769055156b10d0abcd32ef921801","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 13:40:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2975cb2de5434e46baa18e405d6b916d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db0895e8c3a341398afa07184aa4a05d : SUCCESS in 21m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aebbcaca20a74f178858a59ead2692c6 : SUCCESS in 22m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e81568cd221a475587145a82804903a3 : SUCCESS in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ed918cfc8db3d8f4c51d2d80f8698ec5d76b337b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-07 14:00:12.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2340/.","accounts_in_message":[],"_revision_number":4},{"id":"3f74a6a3e8c45b1da1bbb486350de5b9c099f6a3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-07 14:11:20.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7398adf8e35709cebe7e1348b2bf2378f23f893e .","accounts_in_message":[],"_revision_number":4},{"id":"2647c6cbdcf0de1e0a5964f4b3fc8b1f0467281b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-07 14:29:42.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966186 : FAILURE in 1h 10m 38s","accounts_in_message":[],"_revision_number":4},{"id":"83939a1a0083d75ccb295f9e5a8ea0203ce07c20","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-07 14:36:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/40b440b4bba94b6189d3522ced1ab298 : SUCCESS in 1h 05m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0a3ffee155464675a4bc1af2e649e607 : FAILURE in 1h 16m 21s","accounts_in_message":[],"_revision_number":4},{"id":"5bec6628f2a2b4f509663c231724a82b26336945","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-07 14:57:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2454/.","accounts_in_message":[],"_revision_number":4},{"id":"dfda67753f41479f2436c9ad206c88b7c6bc7b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 15:57:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d629ca883e2a4a45ba0965a977efc952\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/782c5e6c8d0c46d78c0886c92b54c409 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae8d4685642248c7bb94b11b768cf4ac : SUCCESS in 10m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02387336db0a4f33a1c1550a41d65a05 : SUCCESS in 15m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/420c5f1911fd44b0a583761bc7692211 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/05251497856849fd8cd7676427b11979 : SUCCESS in 59m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a59c6e63791c4ad7b6e60dc7e519793d : SUCCESS in 1h 02m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/277567817a1b4ebbbdf541b2653c9cd1 : SUCCESS in 1h 48m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26762418521d43d18448c9620c418ec4 : SUCCESS in 1h 06m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e86f86ff141b4e4cb88395eb8e7cd098 : SUCCESS in 16m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6a1db2d60ef548b48814a5338c6fca9f : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/20a2f57855f4450a938aa83b456798ed : SUCCESS in 7m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f9caaa4f4844de3ae50193d89da2c12 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/86fba9ef72e04fde9d8e612a961df6c4 : SUCCESS in 7m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d798a389fde04ed7aed983a74665c15e : SUCCESS in 8m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/aae8cab8fbb3439f9f6d6f47c694fae2 : FAILURE in 45m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/40c4fb84627847bdb5abbdb3fa3d67ea : SUCCESS in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/02317d50fb1d494c85b01f5f9e6d2fe2 : SUCCESS in 1h 57m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/16efaa6bc5be4b5a8c8dab19cc08a56d : FAILURE in 2h 31m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c10457cb82eb464387a3078f0dd2505d : SUCCESS in 1h 59m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a6ab8a74e8e04e5690b355b41d8534e6 : SUCCESS in 2h 05m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/86ec15ca1b23486fa2a208eb759fec3a : SUCCESS in 31m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1bd5635886f7429488fa0ffde9f75ef2 : SUCCESS in 1h 06m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b481ea2561d446b39327e9a2e0a3b2fa : SUCCESS in 1h 36m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1923cb20a5f248d1888c03237878640d : SUCCESS in 42m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6ed2017f467241a284540e9fd94e0e9c : SUCCESS in 2h 11m 07s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0d038011bf8346ee9e93dda23cab3473 : SUCCESS in 2h 32m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ce1a04b93e2d4751a2852bb506b3f0f7 : SUCCESS in 1h 52m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fec3cbddd1984e49ba7b6b7bf425e5c0 : SUCCESS in 58m 24s","accounts_in_message":[],"_revision_number":4},{"id":"04a5c93b93464226542fed4f5d1e7bf3ffc10533","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-11-17 14:42:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a6086346697c43fdec984d6ab5f405fb2c871935","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-17 15:44:57.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7398adf8e35709cebe7e1348b2bf2378f23f893e .","accounts_in_message":[],"_revision_number":4},{"id":"88d804f65028cb5fa6b500d8d3785fe23478cbfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-01-13 13:43:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ace6327f35f2d0f72b44d2c26b4636d3c4b4481d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 14:48:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c169920a973b42119925d5de7810849d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e338869b99c4e558cb093dc90cb8c48 : FAILURE in 14m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79100a09076740d7999782e09fa9fdb3 : FAILURE in 13m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce8e20ea3bdb4ac78d48635abf22ce0a : FAILURE in 16m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7e84078ad2b0f37b46d115a06efd459e5fe12a01","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-13 14:57:08.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7398adf8e35709cebe7e1348b2bf2378f23f893e .","accounts_in_message":[],"_revision_number":5},{"id":"bddff202a14fe4b1d4617fb9ca9e9c6363a1a7fa","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-13 16:40:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/a63555163f1143559313d6bbf7a30cd9 : FAILURE in 26m 26s","accounts_in_message":[],"_revision_number":5},{"id":"89e149f6eddc6ba6584dafe3559e47cac0149f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 16:45:14.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/261c85cea81c4681b223b8408d9d3c87\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4f3301649740feb5705fb18a894340 : FAILURE in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2e2cbf09356426bb173e73a8a3f5c41 : FAILURE in 11m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3eb93e53b2184904a927d4501e704716 : FAILURE in 14m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19f903749ff5485499dd41e7b6422181 : SUCCESS in 13m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/b96b91fc02884e58a7474ccff2c9610f : SUCCESS in 56m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/206160f13be24ed8a2903e6da46c1b7c : SUCCESS in 1h 05m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5722875e39184d069486d335def61f30 : SUCCESS in 1h 45m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5591e7b9b2fb49dfa51d36e0f5c200a7 : SUCCESS in 1h 02m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8fea391a89c240ea9a3cd3afff342b6a : FAILURE in 24m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c6630d708e39479b89227fd799990034 : SUCCESS in 5m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce4e917d2464416c8e8d4aa660b46254 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1973609155b4eecb502abdb6df6f0c9 : SUCCESS in 6m 13s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b3a6f6d96b674e58bf90ef6c44bebfc0 : SUCCESS in 6m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9affde8985e84f34b280937564223e2c : FAILURE in 37m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/31ea5028c8714fc6a1f587dd6a41b25f : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/65e0b716d8684d53835c7f76919b834a : SUCCESS in 2h 09m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/babe3aaef4484d51862bade3e7edcbb2 : FAILURE in 2h 56m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea10404e34e64aacbd1c0de3c6acae70 : SUCCESS in 1h 54m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e80ec9151de74569b99d5b08c3f24f30 : SUCCESS in 1h 57m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6931bda757074708a1de3421ff651a9c : SUCCESS in 35m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91daf4927a0d408297c65c40ef79c698 : SUCCESS in 1h 12m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ba1647a70c674dfa93c17600b48887a2 : SUCCESS in 1h 28m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2c632865d2984ff8b70cacbcee26ff68 : SUCCESS in 32m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cb8781bf0ae44cf1973008834796ec5e : SUCCESS in 1h 08m 33s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9959c700441d42ada5107e837504e5ca : FAILURE in 2h 29m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a606aa20765a4d9f9868243ddb2014c9 : SUCCESS in 1h 42m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc424880d7de429f81ebc0af19279df7 : SUCCESS in 44m 53s","accounts_in_message":[],"_revision_number":5},{"id":"e958a7ec7c59ad05e1040af1f20871be6322d794","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-13 18:06:48.000000000","message":"Patch Set 5:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_200/index.html : SUCCESS in 68m","accounts_in_message":[],"_revision_number":5},{"id":"a732d44add4917efcdad62c32c068555db82218c","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-13 18:06:59.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/39ac508e71c64178aa71f58da7695098 : FAILURE in 32m 00s","accounts_in_message":[],"_revision_number":5},{"id":"36220f0f8bbf3a519e3598d475dec1622b9f8d0b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-13 18:30:41.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966186 : FAILURE in 36m 49s","accounts_in_message":[],"_revision_number":5},{"id":"420e2c8bc73d8ba4c9d6d3e2ba72c56864bb36f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-01-14 13:44:59.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"2cd3cdbc1bc93b057b40a2028c43ce0897d13562","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 13:45:13.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 891602,9","accounts_in_message":[],"_revision_number":6},{"id":"28cd9fd4511828c6d9a48bba6d95c2f2cbf991ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 13:45:13.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 891602,9","accounts_in_message":[],"_revision_number":6},{"id":"4c4c2a591d5242de0a203f329af8bd26583eea88","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-14 13:45:33.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 891602,9","accounts_in_message":[],"_revision_number":6},{"id":"93a84d04e194c6ff89f26e49d37403d6e776badc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-14 13:46:09.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"66cb18c70343a546de794cafe4d8894e7a78ee8e","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-14 13:49:09.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 891602,9","accounts_in_message":[],"_revision_number":6},{"id":"317dd5442255f509821117ee94326b3820b7fa62","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-14 13:49:10.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 891602,9","accounts_in_message":[],"_revision_number":6},{"id":"237d0e623f8e41c48c2daa45de8a5d566f903909","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-14 14:49:42.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I7398adf8e35709cebe7e1348b2bf2378f23f893e .","accounts_in_message":[],"_revision_number":6},{"id":"298beab9242a8cdee634c9e3f12a25806cb45c34","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-14 20:18:03.000000000","message":"Patch Set 6:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_211/index.html : SUCCESS in 71m","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ee1b24f22cf9fe7cc91c9aee6e8f75506a381005","revisions":{"624cc42296cedaf07243176240608a5b76bf608e":{"kind":"REWORK","_number":1,"created":"2025-11-05 14:45:19.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/86/966186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/86/966186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/86/966186/1"}}},"commit":{"parents":[{"commit":"84c24c4ef5e665121a6913b262df6091891e1745","subject":"Support waiting for volume extend completion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84c24c4ef5e665121a6913b262df6091891e1745"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-11-03 13:23:29.000000000","tz":60},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-11-05 14:45:08.000000000","tz":60},"subject":"quobyte: Fix extending attached volumes #1870367","message":"quobyte: Fix extending attached volumes #1870367\n\nFixes extending attached volumes with the Quobyte driver, similar to the\nfix in the NFS driver in change I325d7e54ffd254490938e1effe383eaf1ed0b3bc .\n\nPartial-Bug: #1870367\nChange-Id: I7398adf8e35709cebe7e1348b2bf2378f23f893e\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/624cc42296cedaf07243176240608a5b76bf608e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/624cc42296cedaf07243176240608a5b76bf608e"}]},"branch":"refs/heads/master"},"e21948bf53f727ee6a26ca158ffb995557a8a30d":{"kind":"REWORK","_number":2,"created":"2025-11-05 14:53:48.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/86/966186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/86/966186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/86/966186/2"}}},"commit":{"parents":[{"commit":"84c24c4ef5e665121a6913b262df6091891e1745","subject":"Support waiting for volume extend completion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84c24c4ef5e665121a6913b262df6091891e1745"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-11-03 13:23:29.000000000","tz":60},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-11-05 14:53:35.000000000","tz":60},"subject":"quobyte: Fix extending attached volumes #1870367","message":"quobyte: Fix extending attached volumes #1870367\n\nFixes extending attached volumes with the Quobyte driver, similar to the\nfix in the NFS driver in change I325d7e54ffd254490938e1effe383eaf1ed0b3bc .\n\nPartial-Bug: #1870367\nChange-Id: I7398adf8e35709cebe7e1348b2bf2378f23f893e\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e21948bf53f727ee6a26ca158ffb995557a8a30d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e21948bf53f727ee6a26ca158ffb995557a8a30d"}]},"branch":"refs/heads/master"},"c1ae2d4dbfe792b79d431242e48cd49981f03a2a":{"kind":"REWORK","_number":3,"created":"2025-11-06 13:50:22.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/86/966186/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/86/966186/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/86/966186/3"}}},"commit":{"parents":[{"commit":"84c24c4ef5e665121a6913b262df6091891e1745","subject":"Support waiting for volume extend completion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84c24c4ef5e665121a6913b262df6091891e1745"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-11-03 13:23:29.000000000","tz":60},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-11-06 13:50:11.000000000","tz":60},"subject":"quobyte: Fix extending attached volumes #1870367","message":"quobyte: Fix extending attached volumes #1870367\n\nFixes extending attached volumes with the Quobyte driver, similar to the\nfix in the NFS driver in change I325d7e54ffd254490938e1effe383eaf1ed0b3bc .\n\nPartial-Bug: #1870367\nChange-Id: I7398adf8e35709cebe7e1348b2bf2378f23f893e\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1ae2d4dbfe792b79d431242e48cd49981f03a2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1ae2d4dbfe792b79d431242e48cd49981f03a2a"}]},"branch":"refs/heads/master"},"999a68f0ebd18b3e599ad0071fbc86876cf573b4":{"kind":"REWORK","_number":4,"created":"2025-11-07 13:16:12.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/86/966186/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/86/966186/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/86/966186/4"}}},"commit":{"parents":[{"commit":"84c24c4ef5e665121a6913b262df6091891e1745","subject":"Support waiting for volume extend completion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84c24c4ef5e665121a6913b262df6091891e1745"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-11-03 13:23:29.000000000","tz":60},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-11-07 13:15:59.000000000","tz":60},"subject":"quobyte: Fix extending attached volumes #1870367","message":"quobyte: Fix extending attached volumes #1870367\n\nFixes extending attached volumes with the Quobyte driver, similar to the\nfix in the NFS driver in change I325d7e54ffd254490938e1effe383eaf1ed0b3bc .\n\nPartial-Bug: #1870367\nChange-Id: I7398adf8e35709cebe7e1348b2bf2378f23f893e\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/999a68f0ebd18b3e599ad0071fbc86876cf573b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/999a68f0ebd18b3e599ad0071fbc86876cf573b4"}]},"branch":"refs/heads/master"},"4f47052163978d75f0be6c201909c1c05e8d68f3":{"kind":"REWORK","_number":5,"created":"2026-01-13 13:43:42.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/86/966186/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/86/966186/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/86/966186/5"}}},"commit":{"parents":[{"commit":"7f4f4d68c1105bada3a5b0c48e0663ebaeb088bf","subject":"Quobyte: adapt to image_utils fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f4f4d68c1105bada3a5b0c48e0663ebaeb088bf"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-11-03 13:23:29.000000000","tz":60},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-01-13 13:43:31.000000000","tz":60},"subject":"quobyte: Fix extending attached volumes #1870367","message":"quobyte: Fix extending attached volumes #1870367\n\nFixes extending attached volumes with the Quobyte driver, similar to the\nfix in the NFS driver in change I325d7e54ffd254490938e1effe383eaf1ed0b3bc .\n\nPartial-Bug: #1870367\nChange-Id: I7398adf8e35709cebe7e1348b2bf2378f23f893e\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f47052163978d75f0be6c201909c1c05e8d68f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f47052163978d75f0be6c201909c1c05e8d68f3"}]},"branch":"refs/heads/master"},"ee1b24f22cf9fe7cc91c9aee6e8f75506a381005":{"kind":"REWORK","_number":6,"created":"2026-01-14 13:44:59.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/86/966186/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/86/966186/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966186/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/86/966186/6"}}},"commit":{"parents":[{"commit":"84c24c4ef5e665121a6913b262df6091891e1745","subject":"Support waiting for volume extend completion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84c24c4ef5e665121a6913b262df6091891e1745"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2025-11-03 13:23:29.000000000","tz":60},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-01-14 13:42:25.000000000","tz":60},"subject":"quobyte: Fix extending attached volumes #1870367","message":"quobyte: Fix extending attached volumes #1870367\n\nFixes extending attached volumes with the Quobyte driver, similar to the\nfix in the NFS driver in change I325d7e54ffd254490938e1effe383eaf1ed0b3bc .\n\nPartial-Bug: #1870367\nChange-Id: I7398adf8e35709cebe7e1348b2bf2378f23f893e\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee1b24f22cf9fe7cc91c9aee6e8f75506a381005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee1b24f22cf9fe7cc91c9aee6e8f75506a381005"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
